)]}'
{"id":"openstack%2Fcinder~849803","triplet_id":"openstack%2Fcinder~master~I118339e87eefca2f0df90992c81d4f09fb6d0b68","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-07-15 18:07:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I118339e87eefca2f0df90992c81d4f09fb6d0b68","subject":"HPE3PAR: Correct volume name in ERROR log","status":"MERGED","created":"2022-07-14 10:49:05.000000000","updated":"2022-07-18 09:19:15.000000000","submitted":"2022-07-15 18:07:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849803","meta_rev_id":"7810d9c5bc16404ca655ab7a8cdeebec70f3a305","_number":849803,"virtual_id_number":849803,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-15 18:07:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"recommended":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"all":[{"value":1,"date":"2022-07-15 10:49:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-15 14:27:34.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":2,"date":"2022-07-15 14:25:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-15 14:27:34.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-14 10:56:51.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-07-14 11:15:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-14 11:59:39.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-07-14 13:00:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-14 13:00:36.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-14 13:17:06.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-07-14 13:59:16.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-14 14:18:11.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-14 14:30:55.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-14 14:38:00.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-07-14 14:45:42.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-14 17:12:13.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-14 17:17:05.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-07-14 19:26:37.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-07-15 10:49:00.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2022-07-15 14:25:02.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2022-07-15 14:27:34.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2022-07-18 01:20:16.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"7bed223471d2e475b71c7ae1e010141a056486f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-07-14 10:49:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7865b9617bab1f5074764c3b0d5e83f494f863a5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-14 10:56:51.000000000","message":"Patch Set 1:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/65","accounts_in_message":[],"_revision_number":1},{"id":"03e5ab56d954caaf237193d436c9218e079da9b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 11:15:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b433057dc48497aa713c0200a41fb60\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2cbf5c9c569546d489f6d9aacfa0c4a8 : SUCCESS in 13m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5198cb1c9d59458389683b23e4a473ad : SUCCESS in 23m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a765e6ff41d99c72bc02345a1a6b9ed1aa518536","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-14 11:59:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/849803/1/iscsi/ : SUCCESS in 1h 10m 02s\n- cinder-smb http://cloudbase-ci.com/cinder/849803/1/smb/ : SUCCESS in 57m 15s","accounts_in_message":[],"_revision_number":1},{"id":"5237b84323c51dcae00978755461b56e553d2a07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 13:00:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d6c97efd4774f1d93fb790bd688aca7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f377bfe210743a9962894d0be580b1f : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c9fd72e4eb046ddac346f2afb495fac : SUCCESS in 10m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d126f1b7b4948719402de46e495834f : SUCCESS in 11m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9bdd72f2e544ec6b21d07f9c23b1932 : SUCCESS in 9m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afdf1a9d70474a0cab4b26ca9ba7dbec : SUCCESS in 10m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/d2af85e7172d4584935301959aa02fa4 : SUCCESS in 55m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e40f48f9b823401c9e852aab71d60d7e : SUCCESS in 1h 13m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3054d057c8a94280a564f13f8a839434 : SUCCESS in 47m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/44788d6e29da4831bdac85cf8ef890f5 : SUCCESS in 14m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/20cbdc94dcad41cc96327eccc1ad1c6b : SUCCESS in 4m 35s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/390cb06774044dc6b66212ac4ade0309 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4077866a0d7e493c95d15f675fb00fd1 : SUCCESS in 6m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/339209d7539e40f8888c4c4cb59e4cd8 : SUCCESS in 8m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d432ce0ff0c047019bb092844575d530 : SUCCESS in 40m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7d34b37895a44e4ba68b191c3c8517e9 : FAILURE in 7m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/26d84998dcf64f0dbdb91c62a095d57a : SUCCESS in 1h 27m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a0e2d0ebe425426db514e88ab4b706ed : FAILURE in 1h 36m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3eb7879abb5348918e780e9875e1c41f : SUCCESS in 1h 42m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/00a7c3a624bc4dc0b4c0e3d496c0eb1b : SUCCESS in 1h 26m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8ee23d711bb94f479e7e330bbf2e9a24 : SUCCESS in 1h 52m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0b4a1cacb37240c29ba0dee2f1ff1047 : SUCCESS in 40m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/806688826a9a4c19a7f82d5ae223bfc1 : SUCCESS in 1h 23m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/defca0bfa76f491388ae514f05996d64 : SUCCESS in 2h 09m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d16317a5626e4424bd4e825ce06cacff : SUCCESS in 54m 12s","accounts_in_message":[],"_revision_number":1},{"id":"b0d3b58ae7736b1728902898a6447c7bae6ef47a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-14 13:00:36.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/03/849803/1/check/devstack-plugin-ceph-dsvm-focal-py3/0397429/ : SUCCESS in 1h 46m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d0be72e63e789e07a4b7c8e55892845053b236ee","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-14 13:17:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/03/849803/1/upstream-check/cinder-tempest-plugin-ontap-fcp/73f6756/ : RETRY_LIMIT in 7m 26s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/03/849803/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/e22df5a/ : RETRY_LIMIT in 7m 29s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/03/849803/1/upstream-check/cinder-tempest-plugin-ontap-nfs/e65bd25/ : RETRY_LIMIT in 7m 19s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"80877fef434e0c1829be87620a43b1e03c6d4ab3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-14 13:28:12.000000000","message":"Patch Set 1:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/81","accounts_in_message":[],"_revision_number":1},{"id":"f3cd0465ad8c9ceb1ea93ce2c487e2d790e97712","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-07-14 13:59:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/382c106bfc1f4ce19a173ce571202df1 : SUCCESS in 45m 24s","accounts_in_message":[],"_revision_number":1},{"id":"ca315a1b78a25aada1b6f98d7b458910feb08dd9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-14 14:18:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/849803 : FAILURE in 1h 27m 33s","accounts_in_message":[],"_revision_number":1},{"id":"9cc8eb4c5b9ee1ba67cb4dc983afde6fa9284fb6","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-14 14:30:55.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/03/849803/1/check/fujitsu-eternus-dx-iscsi/b613f66 : SUCCESS in 1h 27m 42s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/03/849803/1/check/fujitsu-eternus-dx-fc/cc8b44f : SUCCESS in 1h 38m 17s","accounts_in_message":[],"_revision_number":1},{"id":"29be831b66c70380732c4975fc8b2706f8c204c4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-14 14:38:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/03/849803/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/372b975/ : SUCCESS in 1h 31m 57s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/03/849803/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a63b080/ : SUCCESS in 1h 27m 06s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/03/849803/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/a5540b9/ : FAILURE in 28m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ce8ce941a6248205ce44d88a70d72280a1fe7579","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-14 14:45:42.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-03-849803-1/ns5_nfs : SUCCESS in 3h 26m 29s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-03-849803-1/ns5_iscsi : SUCCESS in 3h 27m 57s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"7e01e6ecee6ddd9c64556621f8253bfd0427753b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-14 17:12:13.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/03/849803/1/check/DellEMC_PowerFlex/574541b/DellEMC_PowerFlex/991 : FAILURE in 1h 38m 45s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"9bf8ac86125a322ac896990c1e68c4bafb25c3df","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-14 17:17:05.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/03/849803/1/check/dsvm-tempest-cisco-zonemanager-job/287db46/ : FAILURE in 53m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8f07d4a941a05b16d21057af89b8830e2a045b20","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-14 19:26:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/03/849803/1/check/iscsi-driver-nimble/bafda83/bafda83.html : FAILURE in 1h 42m 11s","accounts_in_message":[],"_revision_number":1},{"id":"5489e6c5389cfefe8640c48c5e9baaba23c58180","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-07-15 10:49:00.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fd33a451f89f52ef2a1cdb80a7d5c9563ecb0b26","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-07-15 14:25:02.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf396c04aa393b991d4c3df99f3af4b2c05a2a04","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-07-15 14:27:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2bab3039351a0673f35043c196f8eb436cbcdf2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 14:28:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c2ab1d3887733ca54995157f8ba9916576572e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 18:07:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4057bc4b36264152b12669fa18556414\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1edce6a3598349ffb9bac4f4710e98e9 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dc29a27ffa546b0b46915dc162b13ef : SUCCESS in 13m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8240e97fc2a748f4b4072dc6d72244ff : SUCCESS in 13m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b81d35121c7647bfa07466751dcce372 : SUCCESS in 12m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/6ff1faa2dfb6432d9fe97941883bba18 : SUCCESS in 1h 02m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/051e031d3e6d4363b37ecccb83c9b347 : SUCCESS in 1h 20m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/853cef99c17243f3b25621a04f0052c7 : SUCCESS in 59m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/80ce5080488240dd815238f580bd8808 : SUCCESS in 1h 12m 37s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/012e44c3bbd2478791abd687bf98d3f5 : SUCCESS in 1h 27m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0334788fd7c40a490e4cac45b824130 : SUCCESS in 1h 00m 58s","accounts_in_message":[],"_revision_number":1},{"id":"54d6f94fa0cb81e7c8711b73fb70dcf284c9a015","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 18:07:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1bbc92225e6b87b3a9e107d497da71c4597da984","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 18:08:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b249356aa2941b8b9ed6903222e58a9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f9f5bd25e3944c59027872ccdf253a0 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1},{"id":"3e51887f2fbf72bd0497f716a0c7abd480ab83f7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-18 01:20:16.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"7810d9c5bc16404ca655ab7a8cdeebec70f3a305","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-07-18 09:19:15.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d153ff2621d2af27e889a412724e8495026b4629","revisions":{"d153ff2621d2af27e889a412724e8495026b4629":{"kind":"REWORK","_number":1,"created":"2022-07-14 10:49:05.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/03/849803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/849803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/849803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/849803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/849803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/849803/1"}}},"commit":{"parents":[{"commit":"84fd5986542dcc4a632e6a2f223ef968c1b3aef5","subject":"Merge \"pylint: Add additional ignored-modules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/84fd5986542dcc4a632e6a2f223ef968c1b3aef5"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-07-14 10:09:28.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-07-14 10:09:28.000000000","tz":0},"subject":"HPE3PAR: Correct volume name in ERROR log","message":"HPE3PAR: Correct volume name in ERROR log\n\n3PAR driver creates a dependency chain when we create a volume from\na snapshot and while deleting the snapshot, it renames the dependent\nvolume and also changes some of it\u0027s properties to break the dependency\nto finally delete the snapshot successfully.\nDuring the renaming, 3PAR driver creates a new volume named \"omv-\u003cid\u003e\"\nwhich is different from the original volume name \"osv-\u003cid\u003e\".\n\nNOTE: here \"id\" doesn\u0027t refer to OpenStack resource/volume ID but rather\nthe ID provided by backend.\n\nWe can fail to create the new volume for certain reasons and one of\nthe reason is there already exists an internal volume named \"omv-\u003cid\u003e\"\nSince the IDs are unique, this volume could be a side effect of a\nfailed operation earlier and the volume was not cleaned up.\nThe ERROR log in this failure case shows the original volume name\n\"osv-\u003cid\u003e\" with a Duplicate exception[1] which doesn\u0027t provide any\ninsight of the failure since we already know \"osv-\u003cid\u003e\" exists in\nbackend and we are failing due to \"omv-\u003cid\u003e\" volume which needs to\nbe manually cleaned up from 3PAR array.\nThis patch corrects the volume name in the above failure case.\n\n[1] ERROR oslo_messaging.rpc.server cinder.exception.Duplicate: Volume (osv-\u003cid\u003e) already exists on array.\n\nChange-Id: I118339e87eefca2f0df90992c81d4f09fb6d0b68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d153ff2621d2af27e889a412724e8495026b4629"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d153ff2621d2af27e889a412724e8495026b4629"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
