)]}'
{"id":"openstack%2Fcinder~850251","triplet_id":"openstack%2Fcinder~master~I951179c3abb5b4a40d2ae08765e2a1055f1975c2","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2022-10-14 15:31:56.000000000","reason":"Change was submitted"},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2022-10-14 14:14:17.000000000","reason":"\u003cGERRIT_ACCOUNT_20813\u003e replied on the change","reason_account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}}},"hashtags":[],"change_id":"I951179c3abb5b4a40d2ae08765e2a1055f1975c2","subject":"NFS: Use Volume attr instead of dict compat for lock","status":"MERGED","created":"2022-07-18 19:54:35.000000000","updated":"2022-10-14 15:33:14.000000000","submitted":"2022-10-14 15:31:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850251","meta_rev_id":"0907c597eba0c3d0cb3a5a2d0438b7f3573474d8","_number":850251,"virtual_id_number":850251,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},{"value":0,"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-14 15:31:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-08-19 15:53:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-08-10 06:19:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":1,"date":"2022-08-11 05:24:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},{"value":1,"date":"2022-08-10 06:56:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-14 14:14:17.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},{"value":0,"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-14 14:14:17.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},{"value":0,"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-18 19:55:50.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-07-18 20:17:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-18 22:08:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-18 22:53:42.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-07-18 22:56:15.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-18 23:20:00.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-19 00:20:35.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-19 01:25:56.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-19 01:55:46.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-19 02:42:49.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-07-19 08:08:35.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-07-19 08:38:21.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-07-19 09:07:46.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-07-19 10:02:53.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-20 09:44:11.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-07-24 11:20:40.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-07-26 15:17:03.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2022-07-29 17:25:49.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-10 06:19:30.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2022-08-10 06:56:43.000000000","updated_by":{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},"reviewer":{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},"state":"REVIEWER"},{"updated":"2022-08-11 05:24:08.000000000","updated_by":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"reviewer":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"state":"REVIEWER"},{"updated":"2022-08-19 15:53:06.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"770e74de19332790d8968d9fc33190fcf52637ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-07-18 19:54:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c404e6e661655f43c542b18ab8a75fbaa2487c4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-18 19:55:50.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850251/1/iscsi/ : FAILURE in 44s\n- cinder-smb http://cloudbase-ci.com/cinder/850251/1/smb/ : FAILURE in 45s","accounts_in_message":[],"_revision_number":1},{"id":"07dcffe9555832e027bc2aed285c038d5c752a9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 20:17:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2422285fe8b9476cb74621adec6b6c12\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/240143b8e67f401da606f4f33a9970e1 : SUCCESS in 18m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e8d7066709884f078ccb7ee2082676bc : SUCCESS in 20m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4787ecf746339765ed357ef94004c3c1cacbafb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 22:08:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52345a6f495f43e8bfd11362249e0430\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9982f54fbfd24a1d82dcbc417a81d750 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edef4ed265bd4207bfdd530f9a41ecb8 : SUCCESS in 11m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73cee6731435465e86788fcc7af8e581 : SUCCESS in 11m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78f89d28dbc444139a4c62c4ee474452 : SUCCESS in 11m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1253389ebee406abae739c60f4a9be1 : SUCCESS in 10m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/66866cb9bd754c58848025c1606a0d3e : SUCCESS in 57m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ad8745732b03448182856484f0ddc262 : SUCCESS in 1h 03m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85c1541837914f5594bbad7f4f53c4b6 : SUCCESS in 52m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/558ad209902c48d5bdb64c0810262a42 : SUCCESS in 13m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/eea5fd7f448a42cc9061f813f933fc84 : SUCCESS in 6m 15s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c2b5800f5c0243d5a94646857f12ff2c : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e2571de8129045b7b0a985c42e7d6632 : SUCCESS in 6m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/61ca9253778241599c719f6568395c11 : SUCCESS in 7m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1c55515a67fa483396ada8ec70d16b9f : SUCCESS in 32m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/73ff91767ced42c5ac08dd3d3bd354cb : FAILURE in 8m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f0ad932d821c409ab23f1243d57b24ce : SUCCESS in 1h 26m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/385304501ecd4d2589bf02e635ee377e : FAILURE in 1h 11m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/268d965bcb8f44e6a35ad1cb306ac887 : SUCCESS in 1h 38m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b1f78012d6874c54a2bd088bc896ca27 : SUCCESS in 57m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/17a1f82df91e4a5f8348e8ccb3d8280e : SUCCESS in 1h 21m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/99d319afbd644e9f93494c5c99784c92 : SUCCESS in 32m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3ea63e50aba9433bae830a17231841f6 : SUCCESS in 1h 17m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b8251a2c06b3446382702b86aa05bf45 : SUCCESS in 2h 12m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc20e863384445cebd288027355a92b8 : SUCCESS in 43m 54s","accounts_in_message":[],"_revision_number":1},{"id":"ee40336a661ec9873118d1b2ba5b6e4d89f67018","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-18 22:53:42.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/850251/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/7828d8e/ : SUCCESS in 42m 56s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/850251/1/upstream-check/cinder-tempest-plugin-ontap-nfs/6a7314d/ : SUCCESS in 44m 02s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"942fa95a332720a6b1e1194283864cc70024a42f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-18 22:56:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850251 : FAILURE in 1h 29m 15s","accounts_in_message":[],"_revision_number":1},{"id":"524c19edf666320b43c27b4970b4ad2a0d7feca6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-18 23:20:00.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-51-850251-1/ns5_nfs : SUCCESS in 3h 21m 7s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-51-850251-1/ns5_iscsi : SUCCESS in 3h 24m 54s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"88b89e341c770089dd34a4e6c6a9ad454829267c","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-07-19 00:20:35.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/51/850251/1/check/DellEMC_PowerMAX_FC/a9c7d43/DellEMC_PowerMAX_FC/736 : ABORTED in 10s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/51/850251/1/check/DellEMC_PowerMAX_iSCSI/abaa222/DellEMC_PowerMAX_iSCSI/723 : ABORTED in 18s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"20b43da8fc26514f85405ef4689e79767fbaa281","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-19 01:25:56.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"6d5a6f9c998c28753eb5a24abda090214952fcda","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-19 01:55:46.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/51/850251/1/check/tempest-dsvm-full-focal-py3/227a5e4/ : SUCCESS in 1h 56m 50s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/51/850251/1/check/devstack-plugin-ceph-dsvm-focal-py3/89241b6/ : TIMED_OUT in 3h 08m 03s","accounts_in_message":[],"_revision_number":1},{"id":"4bce11fb5e9919768fdf7ac49d45f682dd7eba03","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-19 02:42:49.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1658174083588 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"3fa1a3ae3fdcba371a069d4a90491a786b57fbad","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-19 08:08:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/51/850251/1/check/dsvm-tempest-cisco-zonemanager-job/bce6066/ : FAILURE in 40m 51s","accounts_in_message":[],"_revision_number":1},{"id":"2713b260d4957db48b5741f4cb4d976568c0013c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-19 08:38:21.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/51/850251/1/check/DellEMC_PowerStore_NFS/dc022d7/DellEMC_PowerStore_NFS/150 : FAILURE in 46m 24s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"7ec40181769e04d7cfbb69598d6d0e91ffd72e6b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-19 09:07:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/51/850251/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/ef9e336/ : SUCCESS in 1h 26m 27s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/51/850251/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/fb72338/ : SUCCESS in 1h 36m 27s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/51/850251/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/2dfd1bb/ : FAILURE in 45m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"894dfeac981443819553942aaa5cf722f6a1796c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-19 10:02:53.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/51/850251/1/check/DellEMC_PowerFlex/3f48cfe/DellEMC_PowerFlex/1023 : SUCCESS in 1h 27m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"dfa13f583856c27695134f36db8aff4b3c717759","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-20 09:44:11.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/51/850251/1/check/DellEMC_Unity_FC/2ab066d/DellEMC_Unity_FC/1188 : SUCCESS in 1h 12m 54s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/51/850251/1/check/DellEMC_Unity_iSCSI/85640da/DellEMC_Unity_iSCSI/1112 : SUCCESS in 1h 57m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"faf37adb67b92b847176643c2759f7ac19e9b026","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-24 11:20:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/51/850251/1/20220724175053/NEC-V-FC : SUCCESS in 1h 55m 39s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/51/850251/1/20220724175053/NEC-V-ISCSI : SUCCESS in 2h 29m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3c7f5b6aa8c864f38f8b66d8bfd0b5974d7fbda0","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-07-29 17:25:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/51/850251/1/20220729225705/NEC-FC : SUCCESS in 3h 28m 03s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/51/850251/1/20220729225705/NEC-ISCSI : SUCCESS in 3h 28m 41s","accounts_in_message":[],"_revision_number":1},{"id":"b104dfa6405c3430e3e804fed9b4768c212153ae","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-08-10 06:19:30.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7280a0a93294b63fd9e3bbab61c62b265c6d0106","author":{"_account_id":32074,"name":"Harsh Ailani","email":"haailani@in.ibm.com","username":"haailani"},"date":"2022-08-10 06:56:43.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"863352f8377b8463a99280924a46d7fe5e36aa66","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2022-08-11 05:24:08.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed51ecbb797bd2cee9e373ee179484c45c965429","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-19 15:53:06.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3d13e3e95a8e2651b492cbc8fcf5b08e208d30d2","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-10-14 14:14:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9f782de06375f14bc240d59872d265d066c15d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 14:15:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"872e0bcd3fbb66309b5a65dd518189e6c642e6ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 15:31:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9981cb0d16a546ac953e2a42ce496ae7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2943730d15941228615e7e1122ebc87 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74181819a3ed4258873bc9f964fdaec0 : SUCCESS in 10m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf6b5e8b57c94b3dbd194cdbc94d0465 : SUCCESS in 9m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6735ecb7cec84896964bc61a44334c86 : SUCCESS in 10m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/45150f6ee30342eeb94062fe4a2101ed : SUCCESS in 54m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5aa1ab06216d4c11a406ddfa55ea4365 : SUCCESS in 1h 12m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47493f6acc134537b7338cdb4d45d387 : SUCCESS in 55m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/34ce1831d59945df94c5f53e710eb5ec : SUCCESS in 1h 03m 40s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2a38289b6ffc4110b96a4e18e524bf06 : SUCCESS in 56m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/762a2c6b913e40e4a2eb53465be2ebf3 : SUCCESS in 42m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d1f203a8040766fec86b682c999232c2a702eee7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 15:31:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0907c597eba0c3d0cb3a5a2d0438b7f3573474d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 15:33:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10b3874fcab84960a8c2b4a916d45103\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/772325c999d3455595c5be9c6116f212 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d402c6fbab6db4cc80e9075f683b80adb2c7368c","revisions":{"d402c6fbab6db4cc80e9075f683b80adb2c7368c":{"kind":"REWORK","_number":1,"created":"2022-07-18 19:54:35.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/51/850251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/51/850251/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/850251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/850251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/51/850251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/51/850251/1"}}},"commit":{"parents":[{"commit":"4754451138bbdccb3917955ef217bd0ce8587f3e","subject":"tests: Address UserWarning in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4754451138bbdccb3917955ef217bd0ce8587f3e"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-07-18 19:51:18.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-07-18 19:51:18.000000000","tz":-240},"subject":"NFS: Use Volume attr instead of dict compat for lock","message":"NFS: Use Volume attr instead of dict compat for lock\n\nUse Volume.id instead of volume[id].\n\nChange-Id: I951179c3abb5b4a40d2ae08765e2a1055f1975c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d402c6fbab6db4cc80e9075f683b80adb2c7368c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d402c6fbab6db4cc80e9075f683b80adb2c7368c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
