)]}'
{"id":"openstack%2Fcinder~850318","triplet_id":"openstack%2Fcinder~master~I08093238df618772f240b0685d98b7a8079e9b81","project":"openstack/cinder","branch":"master","topic":"bp/add-volume-re-image-api","attention_set":{},"removed_from_attention_set":{"34934":{"account":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"last_update":"2023-03-21 05:39:44.000000000","reason":"Change was abandoned"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-08-04 07:05:28.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}}},"hashtags":[],"change_id":"I08093238df618772f240b0685d98b7a8079e9b81","subject":"Add volume status rollback in reimage API","status":"ABANDONED","created":"2022-07-21 02:54:39.000000000","updated":"2023-03-21 05:39:44.000000000","total_comment_count":15,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"fd6c219be7c03551b63310e96f23f75df3fe6e4e","_number":850318,"virtual_id_number":850318,"owner":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-08-15 08:23:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-21 02:56:42.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-07-21 03:23:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-21 04:07:38.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-07-21 04:20:40.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-07-21 04:29:06.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-21 04:32:17.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-07-21 04:58:53.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-07-21 05:10:39.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-21 05:35:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-21 06:20:05.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-21 06:40:23.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-22 02:45:54.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-22 02:52:17.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-07-22 14:58:39.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-07-28 01:07:04.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-08-02 07:39:41.000000000","updated_by":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"reviewer":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"state":"CC"},{"updated":"2022-08-02 12:51:43.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-08-02 16:38:30.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-02 17:26:17.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-08-02 18:18:56.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-08-02 18:49:27.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-03 01:48:26.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-08-03 02:53:36.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-08-04 07:05:28.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2022-08-09 04:22:10.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"3b62759ef5d48ccf0c85ab94863acccee61e36ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"date":"2022-07-21 02:54:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3fd16c16f366e3e1591247b8d0e7f6b1b259121","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"date":"2022-07-21 02:56:40.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"68ca8b9912d7952e7dbe79ce4530d27c51506eec","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-21 02:56:42.000000000","message":"Patch Set 1:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":1},{"id":"326b4b2be5fe43bd48a111aba23a2a63ab998a2d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"date":"2022-07-21 02:57:23.000000000","message":"Topic set to bp/add-volume-re-image-api","accounts_in_message":[],"_revision_number":2},{"id":"90a42d7f27aa859b380618599af1c12ac1b18985","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 03:23:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4de17c1f4a88407bac02dc5047b15d3d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ae702fbdc654ab290316195b4146035 : SUCCESS in 21m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26448321fc4c498abb56cf409862cd22 : SUCCESS in 14m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7cc088b4edd2b53b052c80b9fadcc4a1e09b51dd","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-21 04:07:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850318/2/iscsi/ : SUCCESS in 1h 10m 26s\n- cinder-smb http://cloudbase-ci.com/cinder/850318/2/smb/ : SUCCESS in 1h 00m 49s","accounts_in_message":[],"_revision_number":2},{"id":"ae18d7d0d79e67ff27f598841f9f3472b45399d1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-21 04:20:40.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : RETRY_LIMIT in 0s","accounts_in_message":[],"_revision_number":2},{"id":"405af030878ecf1679cbda138bbd6e56f1e74e0c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-21 04:29:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/18/850318/2/check/fujitsu-eternus-dx-iscsi/f041e44 : FAILURE in 11s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/18/850318/2/check/fujitsu-eternus-dx-fc/128f1d5 : SUCCESS in 1h 23m 41s","accounts_in_message":[],"_revision_number":2},{"id":"836f2f7ca6f50358380bf8813e6d5f7fe1d51df4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-21 04:32:17.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/18/850318/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/9ec26e5/ : SUCCESS in 1h 34m 26s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/18/850318/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/b0991e8/ : SUCCESS in 1h 29m 01s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/18/850318/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/74340ef/ : FAILURE in 32m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5c50a07df09b998a00f754676c0491d5f0013c90","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-07-21 04:58:53.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c44e1ebc40f7be92df5180ee6f04caf648078893","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-21 05:10:39.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/18/850318/2/check/tempest-dsvm-full-focal-py3/4233e7d/ : SUCCESS in 2h 10m 17s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/18/850318/2/check/devstack-plugin-ceph-dsvm-focal-py3/f4bc9c0/ : SUCCESS in 1h 56m 17s","accounts_in_message":[],"_revision_number":2},{"id":"0a83535ce89b2c10da2ac689f5184e96b4033529","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 05:35:11.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b2f44247c31d4b7da1c7bb008c52b908\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a2dc410fd224a1a8bbb9ed7d1023be9 : FAILURE in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50929ee85f1044e6abd5855b9460506a : SUCCESS in 10m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed573b86e5c24046b52ec008df12983b : SUCCESS in 10m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de31bf67e2fc498a954cc9e38e01aeee : SUCCESS in 10m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/168e3ab8eb524aeaab6abc1fe71ceef6 : SUCCESS in 11m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/ae5a1d16184749f5b6a32bf902294cfc : SUCCESS in 51m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9a8751c672f2413eac13d0e576db5ba0 : SUCCESS in 1h 15m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/575fa86a23a741ce92ab30dcfe17c070 : SUCCESS in 46m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3edd951b6f4545ae924b277ef78b4462 : SUCCESS in 17m 12s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5ac9de2f78a94944ad22ccfa9f51d0f5 : FAILURE in 7m 14s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d4ed368f822543e4ba86501f7677ef7a : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0f6b93cb0a2c41edafdce59d8b0a7e4f : SUCCESS in 6m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/718893ac033946799f713f9832ae66d4 : SUCCESS in 6m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fd7b83c66e0a43a1a2368a6ba25b0205 : SUCCESS in 46m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b68eb6cbc5434c0286e12ff02ca86e70 : FAILURE in 6m 43s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/909745c790aa464886b1267f8d5c1925 : SUCCESS in 1h 29m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ea67db8e27b84b5982626dc4a7c57469 : FAILURE in 1h 38m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/847d472037014ba08fc0a0c94b7c3465 : SUCCESS in 1h 58m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5a7b850d4dc84699ba01a3eb7699b1cb : SUCCESS in 1h 22m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1123045368e147aa93b593f9382aa591 : SUCCESS in 1h 39m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/caa563b97b2746788f27da5ade0865f1 : SUCCESS in 55m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bb96e0771c5f4ad392f1748fa79e093e : SUCCESS in 1h 38m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a81f8586848447d8b8b6dc433c3268db : SUCCESS in 2h 29m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64e7de1812774895a4db54cefed37cd1 : SUCCESS in 1h 01m 35s","accounts_in_message":[],"_revision_number":2},{"id":"0201819063a97474e1f6544696f1d622ceb210b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"date":"2022-07-21 05:47:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"673d1dc5b69a763ac5a040f7fc7c9cfd39934ffc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-21 05:47:44.000000000","message":"Patch Set 2:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2},{"id":"87716f1aa93aaec5cba3f722434dc83ca9a952db","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-07-21 06:07:43.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"343fa533a5e0434ea50803b95c9ec1ab95987f64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 06:08:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d637c68da4014a9db3db19a9580c5c3d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0240a9bfc87d4176818169ef26769dee : SUCCESS in 14m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba6913cd5b9b4119b07b3a9251c4910e : SUCCESS in 15m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"827f6381bcf5523bde51d1379bcdec4204b3d410","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-21 06:20:05.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-850318-1/ns5_nfs : SUCCESS in 3h 24m 55s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-850318-1/ns5_iscsi : SUCCESS in 3h 24m 0s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"9875b55bd3b7f513359ff438ff2d04fbe34a01a9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-21 06:32:06.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"5d3f813eead814aaf5873bfdebcf32bbe45b0b0c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-21 06:40:23.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"631ed4bf0fd3091f5c157f6dcafa11fd898ddce9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-21 06:40:43.000000000","message":"Patch Set 3:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":3},{"id":"10ffa6ea73bb6232355ca1b882b94143f44da2ed","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-21 06:44:06.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-850318-2/ns5_nfs : SUCCESS in 3h 26m 22s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-850318-2/ns5_iscsi : SUCCESS in 3h 22m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"36a8a6b132dc00ed6604164b35f0d58c41f6e52a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-21 06:53:23.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850318/3/iscsi/ : SUCCESS in 53m 40s\n- cinder-smb http://cloudbase-ci.com/cinder/850318/3/smb/ : SUCCESS in 1h 05m 08s","accounts_in_message":[],"_revision_number":3},{"id":"ef974e9b20970335164dcbfed17f7fc924e81eb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"date":"2022-07-21 07:23:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2e959ed0185ac5f1ebac9e4432792b106d7e80b5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-21 07:23:08.000000000","message":"Patch Set 3:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":3},{"id":"dc09b1cc659bad40914d7df7cadba8f87c924de4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-21 08:01:51.000000000","message":"Patch Set 4:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":4},{"id":"e88239029334b5c38265d56750522a06f3ebf514","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 08:03:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de7913168d704b4e97a881a809d3e74e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b5d98729ba7a48e9be0a2e0ab48d7919 : SUCCESS in 33m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dfe75cc6b79a4b5da231f0f3c488de7d : FAILURE in 25m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"858f870c3dd4a57b21d0bf28d561845b570610c0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-21 08:03:15.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"e712a9bc02926a643d4b841bfd1fe1802140f5da","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-21 09:10:27.000000000","message":"Patch Set 4:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/18/850318/4/check/fujitsu-eternus-dx-iscsi/0579034 : SUCCESS in 1h 46m 44s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/18/850318/4/check/fujitsu-eternus-dx-fc/a1a93f1 : SUCCESS in 1h 43m 18s","accounts_in_message":[],"_revision_number":4},{"id":"bb879a9892982ff2e8b3fcf7fbafb486a4f79a96","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-21 09:10:59.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/18/850318/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/30cfafc/ : SUCCESS in 1h 46m 08s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/18/850318/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/642efd4/ : SUCCESS in 1h 35m 08s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/18/850318/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/5ef7a5e/ : FAILURE in 31m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4810cde14d58f6361bbfc475a9ebadd906422e01","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-21 09:14:09.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-850318-3/ns5_nfs : SUCCESS in 3h 25m 56s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-850318-3/ns5_iscsi : FAILURE in 10m 19s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"cf3355289d414cb40bd1b2d9fb40c54a58a690ec","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-21 09:19:43.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850318/4/iscsi/ : FAILURE in 1h 34m 35s\n- cinder-smb http://cloudbase-ci.com/cinder/850318/4/smb/ : SUCCESS in 54m 19s","accounts_in_message":[],"_revision_number":4},{"id":"6edcb13a9a5589de9013d614f1cb4efe2f4dc4dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 09:42:05.000000000","message":"Patch Set 4: Verified-1\n\n(11 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d81da2d39c634a5f81c00ae0d218e149\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/327202dbfe4b4d8f88b19d71649cd1b1 : FAILURE in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a345727662f547aabd02e59e958d2fae : SUCCESS in 12m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee32ceb0840f46f085e4def923e4c3bf : SUCCESS in 12m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d7ef94399e8479b98a6d2deccc1bc16 : SUCCESS in 12m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfc5ab916de14fdea5274e266a287ff8 : SUCCESS in 12m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/1e47daefdbd540dcadf684fe3981e00b : SUCCESS in 1h 01m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/17dce2ff8e3c468e96b393e28031ef79 : SUCCESS in 1h 22m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ab15a85c996415a9da640404a7c20be : SUCCESS in 54m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/744c9ea8bb4945018b5432f6d9581732 : SUCCESS in 17m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/759dd772bbc54564929f13272d22a585 : FAILURE in 5m 56s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0fd67149c2b8404f9e71cd6d45f13b9b : SUCCESS in 9m 22s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0f4733555546493289cb833a97c4143d : SUCCESS in 7m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6832f34073304ce7bf0dfa605a4f3765 : SUCCESS in 7m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/12f2b4b7272240fdb6a64a09c645df1a : SUCCESS in 50m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/63f257008df44a1fa9d3ececea7f0a02 : FAILURE in 7m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/92176c7228b24064a8252ea46fc280f7 : SUCCESS in 1h 39m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/887d3e8c89e147d0bfbf20743ef8182d : FAILURE in 1h 55m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/41c01e09ddd54d0ab1e5f56d213d3e89 : SUCCESS in 1h 56m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/962894dfc8cb4e7ca4264d30e55d88e2 : SUCCESS in 1h 24m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/31e83185982346a2bf441e9f080f1b3f : SUCCESS in 1h 37m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/71a23e882c81404d897e0ca7b060c438 : SUCCESS in 44m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/401a59bc028640a5a61509bd372dcc2c : SUCCESS in 1h 10m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a816752888e944c79de055d26a78f609 : SUCCESS in 2h 14m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3d2ae652a5d48cfa1f88b33665e4bc5 : SUCCESS in 46m 34s","accounts_in_message":[],"_revision_number":4},{"id":"6c7955016ca11519ceea7c25599b5129d2a68672","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-21 10:25:08.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/18/850318/4/check/tempest-dsvm-full-focal-py3/afc94a3/ : SUCCESS in 1h 53m 44s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/18/850318/4/check/devstack-plugin-ceph-dsvm-focal-py3/413acbb/ : SUCCESS in 1h 59m 48s","accounts_in_message":[],"_revision_number":4},{"id":"30928e8b637bf73a6b4d6463d931d79f6671ecfe","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-21 10:53:19.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-850318-4/ns5_nfs : SUCCESS in 3h 24m 56s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-850318-4/ns5_iscsi : SUCCESS in 3h 29m 43s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"c671aa5c1254ca5e335ef23444f627cb645c5840","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-21 11:03:25.000000000","message":"Patch Set 4:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/194","accounts_in_message":[],"_revision_number":4},{"id":"408e1d55a0bb22a8bab796bed544f1e0af6bc532","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-07-21 15:05:56.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"25e0c159b95c10c8b01d3feecc0e680eacabd920","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"date":"2022-07-22 01:17:45.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"c6600b43d4845ec0514b5155617f942f3455d696","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-22 01:20:36.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"ef8874a3aa2342d46275553bb01f302b9e8a05d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 01:37:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dcca54b10c142bba679033a816162d4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4392c68019954f1a87fd1d17de5d9138 : SUCCESS in 13m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7745fc0d5f1b4d0fa66d48f97168b3b9 : SUCCESS in 14m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"22d2b0ef8d15e24185ce221ab8b6e717c39499aa","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-22 02:44:55.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850318/5/iscsi/ : SUCCESS in 1h 26m 37s\n- cinder-smb http://cloudbase-ci.com/cinder/850318/5/smb/ : SUCCESS in 1h 04m 54s","accounts_in_message":[],"_revision_number":5},{"id":"3de161f0d1cd272ce38ffe1a9f18ecc8f89d6caf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-22 02:45:54.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850318 : SUCCESS in 1h 26m 39s","accounts_in_message":[],"_revision_number":5},{"id":"a11914257ccdd96e19fc930fb69ed81a56e29078","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-22 02:49:11.000000000","message":"Patch Set 5:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/18/850318/5/check/fujitsu-eternus-dx-iscsi/3271626 : SUCCESS in 1h 29m 26s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/18/850318/5/check/fujitsu-eternus-dx-fc/38b3c33 : SUCCESS in 1h 30m 48s","accounts_in_message":[],"_revision_number":5},{"id":"9339ba04e5bdf972c7f05c5137064fdba7219fb5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-22 02:52:17.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/18/850318/5/check/iscsi-driver-alletra-6k/e861935/e861935.html : FAILURE in 1h 33m 27s","accounts_in_message":[],"_revision_number":5},{"id":"6f96f0c300549c9e76875cd1a98cc95cc7bbdc6b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-22 02:58:12.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/18/850318/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f52f46b/ : SUCCESS in 1h 37m 52s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/18/850318/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/bbe58d6/ : SUCCESS in 1h 39m 02s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/18/850318/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/dd94696/ : FAILURE in 32m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"309f0760c862dd82e01da05dc5976ee4f946c48e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-22 03:40:48.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/18/850318/5/check/tempest-dsvm-full-focal-py3/91b3b02/ : SUCCESS in 1h 59m 20s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/18/850318/5/check/devstack-plugin-ceph-dsvm-focal-py3/b35a44a/ : SUCCESS in 2h 20m 09s","accounts_in_message":[],"_revision_number":5},{"id":"9966b86ed5e5ea83411b70c4711a218952e7c9b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 03:59:50.000000000","message":"Patch Set 5: Verified-1\n\n(11 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ee8456fc1904b55a8cd4ea181de790d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/578b48a42cd74543a520061a56754225 : FAILURE in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/001d3689c8b84f0f87b0cc1a1f597bdd : SUCCESS in 13m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21920a050bb5449b9548502e3ea1ac22 : SUCCESS in 12m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/068feac0b005478cad780929e0d7dc97 : SUCCESS in 14m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/219800d1efba4a378b9ea58a4153ded6 : SUCCESS in 12m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/08c868aab966402aa993a4a6936c80b0 : SUCCESS in 1h 01m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/38cdf7c75b7540fe9ed66b82f180c3cc : SUCCESS in 1h 34m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bea0db2e0a9149b793b2cdca77513f8c : SUCCESS in 53m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a60d66d46c5b492eb16cb5aab0651248 : SUCCESS in 16m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7bd4d30512624b5a8c7fdccc5d0d38db : FAILURE in 6m 04s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/796d6ead38db45b48d101c3fcd32d3c1 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/40c64a2f1d114139a47934c25c1c9bde : SUCCESS in 7m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8154a1561c8a4735a01e2ec51893bd0f : SUCCESS in 7m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/67e4e13344314882b22b483efde2023c : SUCCESS in 46m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/54ce4202d91e46d29c5099d93257c7eb : FAILURE in 6m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ccb9ae7256b44fb991ebf70e844418c3 : SUCCESS in 1h 46m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/77cd0857a4484aca8b25a69c102555a9 : FAILURE in 1h 56m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c9a39d5ea2794a9fa5905b68cb4e7edf : SUCCESS in 1h 48m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6e62b810002b41ff8fe6245465963d5a : SUCCESS in 1h 10m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5367813460584c5a93a08c4965cdb893 : SUCCESS in 1h 38m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/09215649d259499ba6bf50ed18e4bc92 : SUCCESS in 47m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/eab7a6baa7f34250852137b1d19b8740 : SUCCESS in 1h 36m 52s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e7e0485ea1ec4050b831835634ecec23 : SUCCESS in 2h 31m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2399c286a1e41caac0db2a9041189b6 : SUCCESS in 52m 04s","accounts_in_message":[],"_revision_number":5},{"id":"06dc9fc07494dbab09e2b1a425d54f8960f050ce","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-22 04:24:54.000000000","message":"Patch Set 5:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/214","accounts_in_message":[],"_revision_number":5},{"id":"7f1ba0e256dd1c07bd0bc67c00b5d06a998aeed9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-22 04:46:46.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-850318-5/ns5_nfs : SUCCESS in 3h 26m 10s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-850318-5/ns5_iscsi : SUCCESS in 3h 28m 23s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"85c39fb0d61c485f01a34340d1f323ad66188d42","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-22 14:58:39.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/18/850318/5/check/DellEMC_Unity_FC/0be0e39/DellEMC_Unity_FC/1231 : SUCCESS in 1h 10m 57s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/18/850318/5/check/DellEMC_Unity_iSCSI/4626c59/DellEMC_Unity_iSCSI/1135 : SUCCESS in 1h 41m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"39abc47d4a79ccaa77a7e6dcdfb29ad8550f4712","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-28 01:07:04.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/850318/5/20220728075153/NEC-V-FC : SUCCESS in 1h 54m 24s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/850318/5/20220728075153/NEC-V-ISCSI : SUCCESS in 2h 15m 08s","accounts_in_message":[],"_revision_number":5},{"id":"d76267811f360b72bb093c4fb022b4f68439cf1d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"date":"2022-07-28 12:09:25.000000000","message":"Abandoned\n\nAbandon","accounts_in_message":[],"_revision_number":5},{"id":"d09068f7e2a39bcf50b0f565ccd9696099c7d85a","tag":"autogenerated:gerrit:restore","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-08-02 07:38:26.000000000","message":"Restored","accounts_in_message":[],"_revision_number":5},{"id":"9ac384835881672f7d3029909977bc0e9c9fbb05","author":{"_account_id":33771,"name":"IBM Storage CI","email":"katari.kumar@capgemini.com","username":"ibm_svf_ci"},"date":"2022-08-02 07:39:41.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline). Please see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_Storage_CI for more details.\n\n- storwize-tempest-dsvm-full-iscsi storwize-tempest-dsvm-full-iscsi : NODE_FAILURE in 0s\n- storwize-tempest-dsvm-full-fc storwize-tempest-dsvm-full-fc : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"a773d9b0e2d79d663e9d4599cf6919b6e528cb72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 07:57:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab9ed4718b0e4e02ac8c12f002c49eec\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ecdd926578b740cf9846522e1390056b : SUCCESS in 13m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e616aad5405244f6ac717de1576eb706 : SUCCESS in 14m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"414735660753b9c8c2e706d82d570ab4719d3c38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-08-02 08:27:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ada725d57d6281b9c8d814323bccc99e365abca1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-02 08:27:36.000000000","message":"Patch Set 5:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":5},{"id":"371921bd812b23c862671b11614aaa06a4a81726","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-02 08:28:58.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"4ca7cb47c5cb07627fff944ce70536ac4c43efbf","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-08-02 08:31:26.000000000","message":"Patch Set 6:\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4bfc2df01de30c5792bee8ebbd4275bd7677d3b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 08:55:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03afe16943ab4b69a7c7003eca99f730\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/04e176d82617466abbaa86bb46b19d86 : SUCCESS in 23m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d6e10a08cb9a44cea75fbabc3ee80b69 : FAILURE in 25m 17s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"398739eae16358efb76ab0acef863cb2669df8de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 10:33:20.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ddf73b496c44b1cb701b5aa64ce723e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50bc810c62e440639fecf95433746849 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a83428049bc4b3a986fbc34d6090345 : SUCCESS in 11m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/240707dc92d1405fa447d7e2cb245c88 : SUCCESS in 11m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40a67843cc7748b586dc5b938a36a7fa : SUCCESS in 11m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ef85e5ec27244e0921e155451f9fa7f : SUCCESS in 10m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/c47f7f425aa24fe884aab1fc8d876540 : SUCCESS in 1h 11m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6c8a1ce4a81748a3af31bee09c3441b1 : SUCCESS in 1h 25m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac4944fb909a45ff896b712d60a36ea2 : SUCCESS in 1h 01m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b13462d8aedc4861b1bd711289972683 : SUCCESS in 8m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2411ac7630aa4c7497d7862d654634e8 : SUCCESS in 14m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/be86c51ad64b42448668bf2fe34fe86e : FAILURE in 6m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9c028e6c1edc4be9ae258b4e5634327f : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a45bfd67b2d8424c9f68e108d31ce162 : SUCCESS in 5m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/470ed09ecd044d05bd7683e98876840b : SUCCESS in 7m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2a4be58d027348f8acb64ecd1911563e : SUCCESS in 40m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/59b8a75698b540ce8d53e1df69853fb3 : FAILURE in 6m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bd2a636138d14cd5a0468a4995b284b6 : SUCCESS in 1h 09m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f0e796284ff54a688193f3800533d9f4 : FAILURE in 1h 28m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ee753acfb8234039b7c332d247f58110 : SUCCESS in 1h 48m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8043bd2bf6f54b0896656b950c9b35d5 : SUCCESS in 1h 08m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/33f7e12ab74e4e44baadd4a31801f91c : SUCCESS in 1h 40m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/69df9f946b994d2a8e233c37a7c950b5 : SUCCESS in 43m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2744cca568c4414b8b6c092c267b94a4 : SUCCESS in 1h 42m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f57099ee00674abfae478d597aef6879 : SUCCESS in 2h 03m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72a24bc0ffd7412e9662b71638556651 : SUCCESS in 50m 23s","accounts_in_message":[],"_revision_number":6},{"id":"385c3c5467c9ff1e7c6b54c158d1f0c870de9c8d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-02 11:08:49.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850318/6/iscsi/ : SUCCESS in 1h 13m 33s\n- cinder-smb http://cloudbase-ci.com/cinder/850318/6/smb/ : SUCCESS in 56m 53s","accounts_in_message":[],"_revision_number":6},{"id":"9036bc1e352916ed520cc40186dc1c50e9d53059","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-02 11:23:15.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/18/850318/6/check/iscsi-driver-alletra-6k/06ca2ac/06ca2ac.html : FAILURE in 1h 33m 00s","accounts_in_message":[],"_revision_number":6},{"id":"3008ffa54875465a036a4bc502fc866dff6f2255","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-02 12:05:19.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/af6012972ec049a2a34793b28d8c8024 : POST_FAILURE in 2h 17m 33s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/9e9d691b04b64ac8bd14b249d9c65ea8 : POST_FAILURE in 2h 04m 51s","accounts_in_message":[],"_revision_number":6},{"id":"a457f955c8e6cafe0ce7ffc8a49e2685802f303d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-02 12:51:43.000000000","message":"Patch Set 6:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1659428863776 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"e4e7e6df7282f165459f09ce1441061f4cc3c357","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-02 13:05:01.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850318 : FAILURE in 43m 51s","accounts_in_message":[],"_revision_number":6},{"id":"d932b15e6d2d0eff52918ff901d51625ef526bc4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-02 14:10:51.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-850318-6/ns5_nfs : SUCCESS in 3h 28m 0s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-850318-6/ns5_iscsi : SUCCESS in 3h 27m 59s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"4c8280d65b9ce0a64a12f75d80b8c466b2617247","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-08-02 14:33:19.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2088bbbaf3b3e8febffceaf57689315df0efd973","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-02 14:33:21.000000000","message":"Patch Set 6:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":6},{"id":"a5a68c6d88dc70c56bac9c8e3c7c6cd8b9bbc635","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-02 14:34:58.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"64f096a99ea971bbc485eef461e5b2164308491c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 15:05:12.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8bcd785efbb46ca98d722b843057e9c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/85496d035950407bb0b23176427773a1 : SUCCESS in 26m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e7b3df27b47a41e0ab45d0d68a82cb66 : SUCCESS in 29m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d3dc8be67c361d6b0500b52eac96b2d03066402e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-02 15:13:11.000000000","message":"Patch Set 7:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1659450809841 : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"43a57c42077eb0d72d6430bd3359402316e0ef11","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-02 15:47:30.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850318/7/iscsi/ : SUCCESS in 1h 13m 34s\n- cinder-smb http://cloudbase-ci.com/cinder/850318/7/smb/ : SUCCESS in 56m 18s","accounts_in_message":[],"_revision_number":7},{"id":"b404a16e1ad5b0c0b60b15f1a7ec6ef47e84089a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-02 16:06:29.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/18/850318/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f510d1a/ : SUCCESS in 1h 32m 17s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/18/850318/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/fe1359c/ : SUCCESS in 1h 28m 26s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/18/850318/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/d909d6d/ : SUCCESS in 1h 29m 50s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"faa6cdf65ef82cfd1857c70037f827e765c9cc63","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-02 16:11:18.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/18/850318/7/check/iscsi-driver-alletra-6k/6153415/6153415.html : FAILURE in 1h 36m 53s","accounts_in_message":[],"_revision_number":7},{"id":"932cc13f9439993423ccbee9cab227eca866c8f3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-02 16:25:44.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850318 : SUCCESS in 1h 50m 44s","accounts_in_message":[],"_revision_number":7},{"id":"66a033259d2821515220b7bab19dff9145bcda92","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-08-02 16:38:30.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e792f7ed2909479b8796487289209310 : SUCCESS in 43m 11s","accounts_in_message":[],"_revision_number":7},{"id":"862dac7e217d1ae3fd457df2e5e1d66bbdf7689c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 17:07:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1d46d11dc494d9c930c4bf73c064ea2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20197bcc45e24cffa41eb88e47ecc521 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77003fc6aea244438f2f0277132d856d : SUCCESS in 11m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a9b2576e23a43abb49e9f8e8afe3ec9 : SUCCESS in 10m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d88f48c47db48bca9b5a5f11ac63a98 : SUCCESS in 11m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd78554bd39149bab753af1bc5cf4951 : SUCCESS in 11m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/0740b32ab7324df2a08b5866d6c4db27 : SUCCESS in 58m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e5fa518e3a2344cb96518b379d70c1ed : SUCCESS in 1h 24m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/955a130384414d79a2216abb626995f5 : SUCCESS in 57m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cb7c04691b44a40ae54d8e53e077bc7 : SUCCESS in 9m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/886ff997e0bf419e905cebc8f1b27117 : SUCCESS in 18m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3d71f55849054b4a8d091e335c7427d9 : SUCCESS in 6m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f41bfeb37d5a4e4bb77804f5e7657a70 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d629ee0c5b2440f3a37110ff88bd12d3 : SUCCESS in 6m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eb0f383939774c8383028ca2df99a471 : SUCCESS in 7m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a441d1a50f924acc922b27fbbcc14b57 : SUCCESS in 48m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/82c28bb559b24394968b0d9f6bca8ca7 : FAILURE in 6m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/efcfff33f7c54898bd1b225c84d648d2 : SUCCESS in 1h 49m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/80a8358872dd4b3cb88b742519584035 : FAILURE in 1h 57m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a3f29e91b10d46cfa7ba2a242d14c810 : SUCCESS in 1h 52m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9e0608dbf31c439a8d7d5a9ce93ee394 : SUCCESS in 1h 18m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/629555b9778a43f2b5e308a728470ed6 : SUCCESS in 1h 56m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2dc02071e3f74a2d80c0d7ab0e69483c : SUCCESS in 45m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d9a0ab7f7c83483daad22d22634a5fc1 : SUCCESS in 1h 47m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/58acdf4222d6428c8a00b8d8c76fc387 : SUCCESS in 2h 28m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab04b59f354d454eab13eb61b1db17b8 : SUCCESS in 59m 41s","accounts_in_message":[],"_revision_number":7},{"id":"de2d59ab873e3a7dbdc30cb31343de3884d3e6f1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-02 17:11:27.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/68ed367ea1f64394abc6e0c1f74e59d9 : POST_FAILURE in 1h 52m 33s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/25a771cce51f4b5eaac71c329f172fb8 : POST_FAILURE in 2h 30m 00s","accounts_in_message":[],"_revision_number":7},{"id":"606f6a5f136886068bde684ef6054b2cdcc5f7e7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-02 17:26:17.000000000","message":"Patch Set 7:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/02_08_2022_17_07_32\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"15abc16561c1b26d7142ce05249ffad5f25c1784","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-02 18:01:32.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-850318-7/ns5_nfs : SUCCESS in 3h 27m 42s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-850318-7/ns5_iscsi : SUCCESS in 3h 26m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"9b35653418b5a6a0a59d1a75ff973ccac582529a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-02 18:18:56.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/850318/7/upstream-check/cinder-tempest-plugin-ontap-fcp/57ea1db/ : RETRY_LIMIT in 7m 34s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/850318/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/194f127/ : SUCCESS in 1h 05m 31s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/850318/7/upstream-check/cinder-tempest-plugin-ontap-nfs/8a57266/ : SUCCESS in 43m 25s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/850318/7/upstream-check/cinder-tempest-plugin-solidfire-iscsi/9282423/ : RETRY_LIMIT in 7m 26s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"44ac11430b5621b27f20c4cb77ebbe05534cabc4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-02 18:49:27.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/18/850318/7/check/DellEMC_PowerFlex/380ddc5/DellEMC_PowerFlex/1125 : FAILURE in 1h 11m 44s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"6c81a1c14ea5aa1eabc7c129d61214820c142cce","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-03 01:48:26.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/18/850318/7/check/DellEMC_PowerStore_FC/5fbc2e8/DellEMC_PowerStore_FC/664 : FAILURE in 17m 38s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/18/850318/7/check/DellEMC_PowerStore_NFS/d76cc61/DellEMC_PowerStore_NFS/208 : SUCCESS in 1h 28m 41s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/18/850318/7/check/DellEMC_PowerStore_NVMeTCP/907be1c/DellEMC_PowerStore_NVMeTCP/80 : FAILURE in 1h 55m 18s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"8bc09fc5d794b6e8a5b824661ebbe44c9bb93e0f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-08-03 01:50:43.000000000","message":"Patch Set 6:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":6},{"id":"71c8662578cf5decbab8c9528d2600477984d8e8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-08-03 01:50:49.000000000","message":"Patch Set 7:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":7},{"id":"4907561821fd334f0a6a8b5a77dd0b6bdfde2099","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-03 02:13:28.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/18/850318/7/check/fujitsu-eternus-dx-iscsi/2d3d3c8 : FAILURE in 3h 49m 24s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/18/850318/7/check/fujitsu-eternus-dx-fc/f3c4511 : FAILURE in 3h 49m 04s","accounts_in_message":[],"_revision_number":7},{"id":"7c1d763114f9695db511772b8632d3371e4eb372","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-08-03 02:53:36.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/18/850318/7/check/DellEMC_SC_FC/c5cd01f/DellEMC_SC_FC/854 : SUCCESS in 59m 15s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/18/850318/7/check/DellEMC_SC_iSCSI/4f661df/DellEMC_SC_iSCSI/872 : SUCCESS in 2h 06m 16s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"ba68f533fbffa151491c748d414590b520cde39b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-03 07:36:08.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/18/850318/7/check/DellEMC_Unity_FC/b7a25d6/DellEMC_Unity_FC/1340 : FAILURE in 20m 55s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/18/850318/7/check/DellEMC_Unity_iSCSI/d3b5c41/DellEMC_Unity_iSCSI/1218 : FAILURE in 2h 01m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"f0c0d25ebd55a8c914b3ba3aa7eb090abc65499d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-03 14:04:59.000000000","message":"Patch Set 7:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/359","accounts_in_message":[],"_revision_number":7},{"id":"98e044338341af0460dac7aad8322dfa34ce1c71","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-08-03 18:14:13.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/850318/7/20220803204726/NEC-V-FC : SUCCESS in 1h 57m 52s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/850318/7/20220803204726/NEC-V-ISCSI : SUCCESS in 6h 26m 43s","accounts_in_message":[],"_revision_number":7},{"id":"1ece52429656cd4c1e62a2ecda3515433e810b01","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-08-04 07:05:28.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"55217e2e7dc419fe928f77b0180d6c3613156208","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-08-09 04:22:10.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/850318/7/20220809094838/NEC-FC : SUCCESS in 3h 27m 03s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/850318/7/20220809094838/NEC-ISCSI : SUCCESS in 3h 33m 28s","accounts_in_message":[],"_revision_number":7},{"id":"53eb60cb82ac6fc1b90549778c36834d766206bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"date":"2022-08-15 07:30:31.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"e0bf579c4865ffbed2d91b1f85f92cac9c717ad6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-15 07:30:41.000000000","message":"Patch Set 8:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1660548639429 : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"c33ee74b79fc2f026fe82021c057706cc6e09575","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-15 07:44:58.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850318/8/iscsi/ : FAILURE in 13m 51s\n- cinder-smb http://cloudbase-ci.com/cinder/850318/8/smb/ : FAILURE in 13m 54s","accounts_in_message":[],"_revision_number":8},{"id":"9066787b1ab87f507f6856c9f52175999565b24a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 07:45:09.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c456185df600482584524cc047726500\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/326f5d84eaed498098dbc02d424e693b : FAILURE in 6m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/175a440326b04fd787d2a1fd6adf7501 : FAILURE in 7m 41s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"09c33b69075bf671c2375454cadf0b196342ff69","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-15 07:54:25.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/18/850318/8/check/dsvm-tempest-cisco-zonemanager-job/8e92db4/ : POST_FAILURE in 11m 43s","accounts_in_message":[],"_revision_number":8},{"id":"c4d83c5f42127f853c5778a38b657106c411ddb5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-15 07:58:31.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-850318-8/ns5_nfs : FAILURE in 27m 3s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-850318-8/ns5_iscsi : FAILURE in 27m 29s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"1c512e2f6dd07c7762e7f055d42d4e5fe4bd13fc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-15 08:02:09.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/18/850318/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/79dc8f2/ : FAILURE in 27m 17s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/18/850318/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f49870b/ : FAILURE in 30m 25s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/18/850318/8/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/5895737/ : FAILURE in 30m 46s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6daa91487d77608a989285d3ba9089f2d025badb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 08:23:16.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/325c6dd145d54823a78cf6e97f7ebbfa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b9791db28a04589b915a8615a193467 : FAILURE in 8m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7fd522cc99f49caac7d7701edc8bace : FAILURE in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2bb740b085343399298d5586750a3f8 : FAILURE in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9584e26aaaf940799350e760e66577b8 : FAILURE in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9250834ad9e04a53961fa858792a2728 : FAILURE in 6m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/9097cd41b4234634bc09b6984fe151ed : FAILURE in 41m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f42b08f6ac21411faafb2e950b8016a2 : FAILURE in 22m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9176f0f8b96a4662b0f1ab614b07f0e1 : POST_FAILURE in 24m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/363c04d0c9dd4ea3bcb7c06850918097 : SUCCESS in 6m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e4fa8142c82c48e288d4ab24c3dc2607 : FAILURE in 5m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8c3ec227c49d4fc1b15b02016b189fbf : FAILURE in 5m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fd9b7d8e0b5f4ae2a613a9f0d90e449f : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/59c3f0e6face4b33a40b115619f66b76 : FAILURE in 5m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/780dee8ab3034007b6b8b5a1570201c0 : FAILURE in 5m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/08eb8183b34b419eb9eb6fd41028e820 : FAILURE in 22m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a3de05b468664b379766dba680e5c6e4 : FAILURE in 6m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ea34d423241548058412b0374d32d4d0 : FAILURE in 25m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bcc8a17ecdf244c981cee199b0970433 : FAILURE in 27m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fb3956a7d2dd45b0b3e901d5beb70c37 : FAILURE in 40m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d4fa36483127464d80a0b6111e111566 : FAILURE in 50m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ca04367a2f9240c391018a16c9a9afd4 : FAILURE in 35m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/785c0fea056843f4b1c75ddaec84fb39 : FAILURE in 26m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5552778e4de649378a70ea8f8e054345 : FAILURE in 32m 57s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/109a3662c4fe4c2199d863de45d3efe5 : FAILURE in 29m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/307a72197f19466f9d6521b3f60f952b : FAILURE in 34m 56s","accounts_in_message":[],"_revision_number":8},{"id":"403388f39ca94aa3bf968684de2afdba5ca67114","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-15 08:45:29.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/18/850318/8/check/tempest-dsvm-full-focal-py3/464be2f/ : FAILURE in 1h 06m 09s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/18/850318/8/check/devstack-plugin-ceph-dsvm-focal-py3/eaa0dd6/ : FAILURE in 1h 09m 40s","accounts_in_message":[],"_revision_number":8},{"id":"1d105fda5d7f9beb5d1a9ca2773f32533e6f6798","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-15 09:28:11.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/18/850318/8/check/DellEMC_Unity_FC/1f21258/DellEMC_Unity_FC/1454 : FAILURE in 19m 42s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/18/850318/8/check/DellEMC_Unity_iSCSI/fd7ba8b/DellEMC_Unity_iSCSI/1344 : FAILURE in 46m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"a44f8a92116c45cf49c7bb435c1158629fed9d1f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-15 10:32:55.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/18/850318/8/check/iscsi-driver-alletra-6k/60287ed/60287ed.html : TIMED_OUT in 3h 01m 21s","accounts_in_message":[],"_revision_number":8},{"id":"bed108599d7a52da2e2450496ba16198243bb126","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-08-15 12:21:42.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/850318/8/20220815202729/NEC-V-ISCSI : FAILURE in 50m 15s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/850318/8/20220815202729/NEC-V-FC : FAILURE in 54m 10s","accounts_in_message":[],"_revision_number":8},{"id":"83e88cac7838bc6af7bce54a903b0107a3441d60","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-08-21 23:51:00.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/850318/8/20220822080010/NEC-ISCSI : FAILURE in 48m 36s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/850318/8/20220822080010/NEC-FC : FAILURE in 50m 46s","accounts_in_message":[],"_revision_number":8},{"id":"947161023898d999bff52073e4f11c6b093e81f5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-08-29 01:40:29.000000000","message":"Patch Set 8:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":8},{"id":"fd6c219be7c03551b63310e96f23f75df3fe6e4e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"date":"2023-03-21 05:39:44.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"1500bc5f0f95b838e988c6b3c8be93e43e1eff96","revisions":{"2dd44ce4c5fc4856a8f98fdfdd5411a19fab9fa3":{"kind":"REWORK","_number":1,"created":"2022-07-21 02:54:39.000000000","uploader":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"ref":"refs/changes/18/850318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/850318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/850318/1"}}},"commit":{"parents":[{"commit":"7b424fcdcb6cc3cb2d11da04cb6b7744cb13fe20","subject":"Merge \"Add a check for virtual_size at API layer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b424fcdcb6cc3cb2d11da04cb6b7744cb13fe20"}]}],"author":{"name":"yangzhe","email":"17839227138@163.com","date":"2022-07-21 02:52:43.000000000","tz":480},"committer":{"name":"yangzhe","email":"17839227138@163.com","date":"2022-07-21 02:53:55.000000000","tz":480},"subject":" reimage api image  invalid,volume status rollback","message":" reimage api image  invalid,volume status rollback\n\nChange-Id: I08093238df618772f240b0685d98b7a8079e9b81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2dd44ce4c5fc4856a8f98fdfdd5411a19fab9fa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2dd44ce4c5fc4856a8f98fdfdd5411a19fab9fa3"}]},"branch":"refs/heads/master"},"bd47e56ec24ca6ee421d197d981a47814ad14022":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-21 02:56:40.000000000","uploader":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"ref":"refs/changes/18/850318/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/850318/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/850318/2"}}},"commit":{"parents":[{"commit":"7b424fcdcb6cc3cb2d11da04cb6b7744cb13fe20","subject":"Merge \"Add a check for virtual_size at API layer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b424fcdcb6cc3cb2d11da04cb6b7744cb13fe20"}]}],"author":{"name":"yangzhe","email":"17839227138@163.com","date":"2022-07-21 02:52:43.000000000","tz":480},"committer":{"name":"yangzhe","email":"yangshuai09@inspur.com","date":"2022-07-21 02:56:40.000000000","tz":0},"subject":"reimage api image  invalid,volume status rollback","message":"reimage api image  invalid,volume status rollback\n\nCloses-Bug: #1982436\n\nChange-Id: I08093238df618772f240b0685d98b7a8079e9b81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd47e56ec24ca6ee421d197d981a47814ad14022"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bd47e56ec24ca6ee421d197d981a47814ad14022"}]},"branch":"refs/heads/master","description":"Edit commit message"},"aa059119e69c0df27a42624dec0c37ad712a074d":{"kind":"REWORK","_number":3,"created":"2022-07-21 05:47:43.000000000","uploader":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"ref":"refs/changes/18/850318/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/850318/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/850318/3"}}},"commit":{"parents":[{"commit":"7b424fcdcb6cc3cb2d11da04cb6b7744cb13fe20","subject":"Merge \"Add a check for virtual_size at API layer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b424fcdcb6cc3cb2d11da04cb6b7744cb13fe20"}]}],"author":{"name":"yangzhe","email":"17839227138@163.com","date":"2022-07-21 02:52:43.000000000","tz":480},"committer":{"name":"yangzhe","email":"17839227138@163.com","date":"2022-07-21 05:47:28.000000000","tz":480},"subject":" reimage api image  invalid,volume status rollback","message":" reimage api image  invalid,volume status rollback\n\nChange-Id: I08093238df618772f240b0685d98b7a8079e9b81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aa059119e69c0df27a42624dec0c37ad712a074d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aa059119e69c0df27a42624dec0c37ad712a074d"}]},"branch":"refs/heads/master"},"eba6fe5c4e28eb551b5137073b0b37cc75b1bab8":{"kind":"REWORK","_number":4,"created":"2022-07-21 07:23:07.000000000","uploader":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"ref":"refs/changes/18/850318/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/850318/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/850318/4"}}},"commit":{"parents":[{"commit":"7b424fcdcb6cc3cb2d11da04cb6b7744cb13fe20","subject":"Merge \"Add a check for virtual_size at API layer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b424fcdcb6cc3cb2d11da04cb6b7744cb13fe20"}]}],"author":{"name":"yangzhe","email":"17839227138@163.com","date":"2022-07-21 02:52:43.000000000","tz":480},"committer":{"name":"yangzhe","email":"17839227138@163.com","date":"2022-07-21 07:22:54.000000000","tz":480},"subject":" reimage api image  invalid,volume status rollback","message":" reimage api image  invalid,volume status rollback\n\nChange-Id: I08093238df618772f240b0685d98b7a8079e9b81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eba6fe5c4e28eb551b5137073b0b37cc75b1bab8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eba6fe5c4e28eb551b5137073b0b37cc75b1bab8"}]},"branch":"refs/heads/master"},"4e90dd49ebef001a2a401b61d00fac98256800c5":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-07-22 01:17:45.000000000","uploader":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"ref":"refs/changes/18/850318/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/850318/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/850318/5"}}},"commit":{"parents":[{"commit":"7b424fcdcb6cc3cb2d11da04cb6b7744cb13fe20","subject":"Merge \"Add a check for virtual_size at API layer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b424fcdcb6cc3cb2d11da04cb6b7744cb13fe20"}]}],"author":{"name":"yangzhe","email":"17839227138@163.com","date":"2022-07-21 02:52:43.000000000","tz":480},"committer":{"name":"yangzhe","email":"yangshuai09@inspur.com","date":"2022-07-22 01:17:45.000000000","tz":0},"subject":"Add volume status rollback in reimage API","message":"Add volume status rollback in reimage API\n\nCloses-Bug: #1982436\n\nChange-Id: I08093238df618772f240b0685d98b7a8079e9b81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e90dd49ebef001a2a401b61d00fac98256800c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4e90dd49ebef001a2a401b61d00fac98256800c5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"064cbadafa54e7c74cac1c1c3fc4f648dfaab296":{"kind":"REWORK","_number":6,"created":"2022-08-02 08:27:35.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/18/850318/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/850318/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/850318/6"}}},"commit":{"parents":[{"commit":"7b424fcdcb6cc3cb2d11da04cb6b7744cb13fe20","subject":"Merge \"Add a check for virtual_size at API layer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b424fcdcb6cc3cb2d11da04cb6b7744cb13fe20"}]}],"author":{"name":"yangzhe","email":"17839227138@163.com","date":"2022-07-21 02:52:43.000000000","tz":480},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-08-02 08:24:06.000000000","tz":330},"subject":"Add volume status rollback in reimage API","message":"Add volume status rollback in reimage API\n\nWhen we perform reimage operation, we check the image metadata\nlike we do when creating a bootable volume from image.\nIf the metadata check fails for any reason like image status is\nnot active or image size or virtual size is bigger than volume\nsize, we fail the operation but the volume stays in downloading\nstate.\nThis patch adds the handling when the image metadata check fails\nwe revert the volume status to it\u0027s original status.\n\nCloses-Bug: #1982436\n\nCo-Authored-By: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\n\nChange-Id: I08093238df618772f240b0685d98b7a8079e9b81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/064cbadafa54e7c74cac1c1c3fc4f648dfaab296"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/064cbadafa54e7c74cac1c1c3fc4f648dfaab296"}]},"branch":"refs/heads/master"},"36fab54725ca03a6dbb814678798302381307ce3":{"kind":"REWORK","_number":7,"created":"2022-08-02 14:33:19.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/18/850318/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/850318/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/850318/7"}}},"commit":{"parents":[{"commit":"92807d6ceee8847a2cf5a1321cf4e1c9bfdb98f4","subject":"Merge \"backup/swift: Add support sending service user token\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/92807d6ceee8847a2cf5a1321cf4e1c9bfdb98f4"}]}],"author":{"name":"yangzhe","email":"17839227138@163.com","date":"2022-07-21 02:52:43.000000000","tz":480},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-08-02 14:33:06.000000000","tz":0},"subject":"Add volume status rollback in reimage API","message":"Add volume status rollback in reimage API\n\nWhen we perform reimage operation, we check the image metadata\nlike we do when creating a bootable volume from image.\nIf the metadata check fails for any reason like image status is\nnot active or image size or virtual size is bigger than volume\nsize, we fail the operation but the volume stays in downloading\nstate.\nThis patch adds the handling when the image metadata check fails\nwe revert the volume status to it\u0027s original status.\n\nCloses-Bug: #1982436\n\nCo-Authored-By: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\n\nChange-Id: I08093238df618772f240b0685d98b7a8079e9b81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/36fab54725ca03a6dbb814678798302381307ce3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/36fab54725ca03a6dbb814678798302381307ce3"}]},"branch":"refs/heads/master"},"1500bc5f0f95b838e988c6b3c8be93e43e1eff96":{"kind":"REWORK","_number":8,"created":"2022-08-15 07:30:31.000000000","uploader":{"_account_id":34934,"name":"yangzhe","email":"yangshuai09@inspur.com","username":"yangzhe"},"ref":"refs/changes/18/850318/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/850318/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/850318/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/850318/8"}}},"commit":{"parents":[{"commit":"92807d6ceee8847a2cf5a1321cf4e1c9bfdb98f4","subject":"Merge \"backup/swift: Add support sending service user token\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/92807d6ceee8847a2cf5a1321cf4e1c9bfdb98f4"}]}],"author":{"name":"yangzhe","email":"17839227138@163.com","date":"2022-07-21 02:52:43.000000000","tz":480},"committer":{"name":"yangzhe","email":"yangshuai09@inspur.com","date":"2022-08-15 07:30:31.000000000","tz":0},"subject":"Add volume status rollback in reimage API","message":"Add volume status rollback in reimage API\n\nWhen we perform reimage operation, we check the image metadata\nlike we do when creating a bootable volume from image.\nIf the metadata check fails for any reason like image status is\nnot active or image size or virtual size is bigger than volume\nsize, we fail the operation but the volume stays in downloading\nstate.\nThis patch adds the handling when the image metadata check fails\nwe revert the volume status to it\u0027s original status.\n\nCloses-Bug: #1982436\n\nCo-Authored-By: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\n\nChange-Id: I08093238df618772f240b0685d98b7a8079e9b81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1500bc5f0f95b838e988c6b3c8be93e43e1eff96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1500bc5f0f95b838e988c6b3c8be93e43e1eff96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
