)]}'
{"id":"openstack%2Fcinder~850831","triplet_id":"openstack%2Fcinder~master~Ia6dcf8ee2205d32714849c0730c9a3562fa8f485","project":"openstack/cinder","branch":"master","topic":"bp/hitachi-vsp-support-dedup-comp","attention_set":{},"removed_from_attention_set":{"33473":{"account":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"last_update":"2023-02-25 01:02:38.000000000","reason":"Change was submitted"},"28403":{"account":{"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"},"last_update":"2022-09-06 06:20:27.000000000","reason":"\u003cGERRIT_ACCOUNT_28403\u003e replied on the change","reason_account":{"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"}},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2023-02-25 01:02:38.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-02-23 03:53:58.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"Ia6dcf8ee2205d32714849c0730c9a3562fa8f485","subject":"Hitachi: support data deduplication and compression","status":"MERGED","created":"2022-07-25 04:48:01.000000000","updated":"2023-03-01 08:05:21.000000000","submitted":"2023-02-25 01:02:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":39,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850831-bp/hitachi-vsp-support-dedup-comp","meta_rev_id":"3b104ec734c32dd8c3fe6f575d42f61552b5f541","_number":850831,"virtual_id_number":850831,"owner":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"},{"value":0,"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-25 01:02:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":22879,"name":"Prabhuraj K","email":"k.prabhuraj@in.ibm.com","username":"pk1218"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2023-02-23 03:45:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"},{"value":0,"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":2,"date":"2023-02-23 03:53:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":22879,"name":"Prabhuraj K","email":"k.prabhuraj@in.ibm.com","username":"pk1218"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"},{"value":0,"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":1,"date":"2023-02-23 03:53:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":22879,"name":"Prabhuraj K","email":"k.prabhuraj@in.ibm.com","username":"pk1218"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2023-02-23 03:45:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"},{"value":0,"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":22879,"name":"Prabhuraj K","email":"k.prabhuraj@in.ibm.com","username":"pk1218"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22879,"name":"Prabhuraj K","email":"k.prabhuraj@in.ibm.com","username":"pk1218"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-25 04:54:11.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-07-25 05:11:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 05:45:46.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-07-25 05:54:52.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-07-25 06:14:52.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-07-25 06:21:15.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 06:25:40.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 06:46:37.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-07-25 07:05:40.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 07:19:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-25 08:03:06.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-07-25 08:25:36.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 08:27:21.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-07-25 08:35:54.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 09:51:35.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 11:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2022-07-25 11:10:29.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-07-28 13:46:40.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-08-04 00:42:47.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-26 07:42:09.000000000","updated_by":{"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"},"reviewer":{"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"},"state":"REVIEWER"},{"updated":"2022-09-05 15:09:06.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-09-07 06:07:35.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-13 08:33:09.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-21 09:13:46.000000000","updated_by":{"_account_id":22879,"name":"Prabhuraj K","email":"k.prabhuraj@in.ibm.com","username":"pk1218"},"reviewer":{"_account_id":22879,"name":"Prabhuraj K","email":"k.prabhuraj@in.ibm.com","username":"pk1218"},"state":"REVIEWER"},{"updated":"2023-01-03 08:50:05.000000000","updated_by":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"reviewer":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"state":"REVIEWER"},{"updated":"2023-02-09 08:01:54.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-02-09 08:15:58.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-09 17:05:04.000000000","updated_by":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"reviewer":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"state":"CC"},{"updated":"2023-02-09 18:47:46.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-02-09 23:22:37.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2023-02-10 05:29:58.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-02-17 14:49:20.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2023-02-17 14:51:41.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-02-22 12:18:22.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-02-23 03:45:16.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"e00411b306a6707fd07f2cbcc29b52eaf074a3bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-07-25 04:48:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccb06a1d75d1715256c4cd985a70863df7608811","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-25 04:54:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"e17c7a6a77a4b4c6ae2767774b7531918c9ceebc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 05:11:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/938d708dec4147dda8a2ded78398ef36\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8337db69355c4af2b813b120f9a17964 : SUCCESS in 16m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/131a1be32e4e42998a6f1273a502f8c3 : SUCCESS in 18m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a577b43dfa441bf89633d8b5495481588a57d32e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-25 05:45:46.000000000","message":"Patch Set 1:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/226","accounts_in_message":[],"_revision_number":1},{"id":"4718481d47101f44dc4d68cd8c2c5bb7213d8e03","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-25 05:54:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/850831/1/check/iscsi-driver-alletra-6k/260d5eb/260d5eb.html : FAILURE in 1h 05m 49s","accounts_in_message":[],"_revision_number":1},{"id":"7e32aec89de0b4dfe41b1414bfe33e52548475b8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-25 06:14:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850831/1/iscsi/ : SUCCESS in 1h 26m 20s\n- cinder-smb http://cloudbase-ci.com/cinder/850831/1/smb/ : SUCCESS in 1h 13m 08s","accounts_in_message":[],"_revision_number":1},{"id":"adca67d78a3d98d51030ab722ba715e2c1a6c1b2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-25 06:21:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850831 : SUCCESS in 1h 28m 28s","accounts_in_message":[],"_revision_number":1},{"id":"a583bcc53f69a55d69a9bf5e61d120ae43509932","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-25 06:25:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/31/850831/1/check/fujitsu-eternus-dx-iscsi/40d3e74 : FAILURE in 1h 35m 13s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/31/850831/1/check/fujitsu-eternus-dx-fc/78cf534 : FAILURE in 11s","accounts_in_message":[],"_revision_number":1},{"id":"efb88a0916fa427a5c7b4809f97b414cff2922bf","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-25 06:46:37.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/850831/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/c3a661c/ : SUCCESS in 1h 57m 46s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/850831/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6679bb3/ : SUCCESS in 1h 38m 17s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/31/850831/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/da8a379/ : FAILURE in 27m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26b65505f614ec37ba29ba61c290a295fea748f3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-25 07:05:40.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/58da5ebd6ea447feb932959f632c4f2c : POST_FAILURE in 2h 15m 15s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/5902834ea1b941f987b2562c4175880d : POST_FAILURE in 2h 11m 15s","accounts_in_message":[],"_revision_number":1},{"id":"003eae5b4f95b9f3d4a80b05c182c5589ed9f00c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 07:19:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29e23245e3214d9a94851f1b09424333\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de900764853e48b4bc322e0592582174 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f61d91c645294d5cbfc2cdad0c96dfdf : SUCCESS in 10m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4135a98469af427088292c01cf5d7ead : SUCCESS in 13m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72fd69657df44ce7b6091d2d2aa51114 : SUCCESS in 13m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e24aca10b234229892934d122013310 : SUCCESS in 11m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/c36c5bd0abcb46d98714bb3cda055733 : SUCCESS in 1h 08m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fea294ffbf09486fb518f2c59a51111f : SUCCESS in 1h 16m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ed132264f3b4033a7e24031586920e1 : SUCCESS in 59m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd54fb2698084662bdeed9a46c1772de : SUCCESS in 7m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4718d91c9fdb49a4b75e94d731c42459 : SUCCESS in 16m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6996b8ee301c43769ca32a4adb3ce56d : FAILURE in 4m 52s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/34b7daf2ed144b4b820906804280a5f4 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e5edabbab8724b77a1c81958716d4bbf : SUCCESS in 5m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/55bfaff6c946489b82644aa791f0bdf4 : SUCCESS in 6m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/745ead8657074926ad552405b252054c : SUCCESS in 47m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fbcac6e645fb4e8da0e464600ebd5f00 : FAILURE in 5m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2b4edcbf9aa94a47a31db8c659d5354d : SUCCESS in 1h 22m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/49519913a8d94197b019dc64cdee4dc4 : FAILURE in 1h 30m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/062454ea6a8641e3b780c5895d4249c2 : SUCCESS in 1h 38m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/20326ee005f348c4aa00ad3640be3a22 : SUCCESS in 1h 17m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/00f299791c2a4b22a75385ba274e6324 : SUCCESS in 1h 25m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/262d2550f4944f798a6404df2c1b9dbc : SUCCESS in 49m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bbd9c0f57a0b4b54a5b326bcf3f5628c : SUCCESS in 1h 24m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a8ebb462dac54d399f221452b4083924 : SUCCESS in 2h 23m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb483278100a4d1bbd741688a23bbbd7 : SUCCESS in 1h 00m 17s","accounts_in_message":[],"_revision_number":1},{"id":"bd2a670a4b16f3fbcccfabde8b2d694da55d59b5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-25 08:03:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/850831/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/e037fb7/ : FAILURE in 43m 08s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/850831/1/upstream-check/cinder-tempest-plugin-ontap-nfs/d304d23/ : RETRY_LIMIT in 7m 43s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"f0e584ef26d36bc5f2a274d08ed9222b2d04e8ae","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-25 08:25:36.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"1d00c3cf4741d8961a8b3cd6043746d57e1a0129","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-25 08:27:21.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/850831/1/check/DellEMC_Unity_FC/995346e/DellEMC_Unity_FC/1240 : FAILURE in 20m 40s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/850831/1/check/DellEMC_Unity_iSCSI/e8c8cc3/DellEMC_Unity_iSCSI/1144 : SUCCESS in 1h 49m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"355636a99b1003156506957cd845db5d98a0b1a9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-25 08:35:54.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-850831-1/ns5_nfs : SUCCESS in 3h 22m 52s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-850831-1/ns5_iscsi : SUCCESS in 3h 26m 56s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"8b4e6cc6d0a3a7a0d7c4c2999879b8a9100ca71c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-25 09:51:35.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/31/850831/1/check/DellEMC_PowerFlex/ad615da/DellEMC_PowerFlex/1062 : FAILURE in 1h 31m 16s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"a9a044ea136c0986553faf23b421abe7c800a2ed","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2022-07-25 11:00:06.000000000","message":"Patch Set 1:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/850831/1/fc : FAILURE in h m 0s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/850831/1/iscsi : FAILURE in h m 0s","accounts_in_message":[],"_revision_number":1},{"id":"2ede0da4a508596a10cb30f1e9a195d16e957281","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-07-25 11:10:29.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/31/850831/1/check/DellEMC_SC_FC/390029b/DellEMC_SC_FC/795 : SUCCESS in 47m 39s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/31/850831/1/check/DellEMC_SC_iSCSI/537c9aa/DellEMC_SC_iSCSI/818 : SUCCESS in 2h 06m 49s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"8925ef147f63e0f5f308e54671b793e16119a4f6","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-28 13:46:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/850831/1/20220728194744/NEC-V-FC : SUCCESS in 2h 01m 36s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/850831/1/20220728194744/NEC-V-ISCSI : SUCCESS in 2h 58m 52s","accounts_in_message":[],"_revision_number":1},{"id":"551b4be30cf015c754d01b4214714c6387d54f81","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2022-07-29 10:00:07.000000000","message":"Patch Set 1:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/850831/1/fc : SUCCESS in 1h 05m 40s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/850831/1/iscsi : SUCCESS in 1h 28m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0b60340a0bc1523332f724bc50378bf48848f1ec","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-08-04 00:42:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/850831/1/20220804061314/NEC-FC : SUCCESS in 3h 28m 51s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/850831/1/20220804061314/NEC-ISCSI : SUCCESS in 3h 29m 29s","accounts_in_message":[],"_revision_number":1},{"id":"b8bb2bbdf70a4e55bbab60554c7dd3b8574b8540","author":{"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"},"date":"2022-08-26 07:42:09.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"112c287df032f9932d5a4a8662522c40552e716b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-09-02 04:43:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f7c005e9079870c21196054f34adef065dc1fc05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-09-02 04:45:16.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"cfd28e621e1501566ea81d12fcd1a39587992fe2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-02 04:45:17.000000000","message":"Patch Set 2:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2},{"id":"fc334797914ab1703cc520815ed5d92d141684ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 05:08:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7325227874dc4ca4b6108153cff55b8f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4bb72d195c24145b8099d972f9aefe4 : SUCCESS in 21m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c9cbcc5a9c04479495de5791896a511f : SUCCESS in 20m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1538ea1b8956df804ee0e4022ee03e408a729bc6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-02 05:31:00.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/850831 : FAILURE in 42m 08s","accounts_in_message":[],"_revision_number":3},{"id":"839fe61aef4b095ec76542fa46bec2c49ced42c7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-09-02 05:49:19.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850831/3/iscsi/ : SUCCESS in 1h 03m 22s\n- cinder-smb http://cloudbase-ci.com/cinder/850831/3/smb/ : SUCCESS in 53m 15s","accounts_in_message":[],"_revision_number":3},{"id":"b51f26db1553acb5816a64db6620aec09b19b1f7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-09-02 05:54:04.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/850831/3/check/iscsi-driver-alletra-6k/58c06a9/58c06a9.html : FAILURE in 1h 05m 00s","accounts_in_message":[],"_revision_number":3},{"id":"f87f257111574e450a9ac7b90087c47636373b04","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-02 06:29:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850831 : SUCCESS in 1h 32m 06s","accounts_in_message":[],"_revision_number":3},{"id":"740519021c279fc10763a31ccd959f7464ff9e91","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-02 06:59:59.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/850831/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/7252624/ : SUCCESS in 1h 33m 38s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/850831/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/e9b335d/ : SUCCESS in 1h 29m 16s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/31/850831/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/603bd13/ : SUCCESS in 2h 07m 51s","accounts_in_message":[],"_revision_number":3},{"id":"a927a8ac296496b33be14dd226d069dcb57ffe61","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-02 07:01:23.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/850831/3/check/DellEMC_Unity_FC/1ef50b1/DellEMC_Unity_FC/1674 : SUCCESS in 1h 09m 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/850831/3/check/DellEMC_Unity_iSCSI/e9b80db/DellEMC_Unity_iSCSI/1534 : FAILURE in 1s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"466012f29d050a6fe20449565ea0b1b76eea5172","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 07:08:44.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/20ef0470d5fb4426b8cda9f779faa100\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41c82a49c9de4b5b9db50999e6f78329 : FAILURE in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73042d311c5241b6b01e352d374d03e1 : SUCCESS in 11m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a41f8af35fb4a9f9c8917bee8124eea : TIMED_OUT in 40m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6bafef610c8b42cf8e5468e6c56996bf : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c99b4ad6e35444792eedd395e64b356 : SUCCESS in 12m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/3553f57852724399a0960355eea2adac : SUCCESS in 1h 05m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/410f030f378b48d2be5d8b2416e17c55 : SUCCESS in 1h 21m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3e55cad4758449b9312ce9523223f6d : SUCCESS in 1h 13m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11c107c818254fb2bffabc33881437ce : SUCCESS in 8m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e7764c94c9b345199e6126b7e035cc8f : SUCCESS in 15m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a53bc879922c4c65be641a95e73b94a9 : SUCCESS in 5m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1a9cf5eeadb1420dac7016d9912ecd1e : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0184ddb77a324dfd81b685d40dcffb72 : SUCCESS in 6m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3b2ef37efd694e128942ac5e5f67b593 : SUCCESS in 9m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2200f9ac257b4797b80a569c837cb602 : SUCCESS in 49m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/73855fd4745b457e95f9f5b8798c2370 : FAILURE in 7m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b324670a4f0145f9a65b9e5befc82636 : FAILURE in 1h 31m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8a757193919e464693e3951e4c7b95f9 : FAILURE in 1h 41m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ea3822563770476f838f3c340dfdc2d3 : SUCCESS in 1h 58m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ee40ea23db554ebba54314f6a7eea72c : SUCCESS in 1h 18m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3db27aa8d8b34f63ad22c02c3423dae9 : SUCCESS in 1h 29m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b49b2bfd0d3645749788a47fade3b820 : SUCCESS in 45m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b7609c078c764fd990bc9b2e6ea35874 : SUCCESS in 1h 38m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1250f20493e04be2b27889055223de3a : SUCCESS in 2h 16m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3efce8c3b9a4d4e9161ccfc4a43989e : SUCCESS in 58m 53s","accounts_in_message":[],"_revision_number":3},{"id":"7098cfa11b1382c7ae597d16e5551ee0d21f744a","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-09-02 08:39:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f6472fea3889a2baf95d8b39bf63ad98f89dd8ee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-02 08:39:21.000000000","message":"Patch Set 3:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":3},{"id":"1f10f31a01cd60868a6fa767e99ca5e606f71876","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-02 08:55:40.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-850831-3/ns5_nfs : SUCCESS in 3h 29m 47s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-850831-3/ns5_iscsi : SUCCESS in 3h 28m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"1725886d5d8fbdeea8379982dbc3ea5b8353f11c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 09:00:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82c799c7c8364d91b40b020ffe7c93bb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c921047fb5844cd803e9c34e7748fbf : SUCCESS in 12m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0d530b0068f549378a291b88ee3798cd : SUCCESS in 19m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5a591281f0e144761f5fea2df6c42690fd176f64","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-02 09:05:12.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-850831-2/ns5_nfs : FAILURE in 4h 1m 58s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-850831-2/ns5_iscsi : SUCCESS in 3h 31m 57s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"5e3bc45bdedb0e6948b4025f9411c33d3302e80a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-09-02 10:25:14.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850831/3/iscsi/ : FAILURE in 1h 44m 52s\n- cinder-smb http://cloudbase-ci.com/cinder/850831/3/smb/ : SUCCESS in 1h 04m 40s","accounts_in_message":[],"_revision_number":3},{"id":"51ea679a1c10e5f8e12b52853e61f49111f29347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 10:50:08.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a938742bc69a458ab9e75593bb8103d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f6607664f6e4a92933a6a6cf26c945c : FAILURE in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/892b2e39e09d44279caa66f6e51dcfef : SUCCESS in 14m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa46052277f4406f9c997c97338a9bdf : SUCCESS in 11m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4370ba66ab1f4c81a898e0e655da4aff : SUCCESS in 12m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d34af501234b45839b156ec91265b344 : SUCCESS in 14m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/e9ba16298b3d43c481b10a95bda439b9 : SUCCESS in 59m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d01259e5ed3a4fb98cbf51fe6f3204ab : SUCCESS in 1h 26m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c575f2e359f844e08b49af94684d579e : SUCCESS in 1h 09m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/318b369dde2e479090214f5f6783e02a : SUCCESS in 9m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3719b8a8e6d14964bc6fcab48f059db6 : SUCCESS in 16m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/830fd6caa13f4acbbab3cf4cd87303af : SUCCESS in 6m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6cbb2e6b68f74a75896d2c5f84c3775b : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8184496611344dc1b5a2ff49c0c487c2 : SUCCESS in 8m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bbf725b67ee44732b2624770ead84140 : SUCCESS in 6m 10s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/247578970d6047f2896f004dd58cf91e : SUCCESS in 46m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3f09f8f079ac48479926b60d3d009e14 : FAILURE in 8m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a79e1c75d04b481094dad347d69ea843 : SUCCESS in 1h 27m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/38624ba38b1c426a98a10d67417162ed : FAILURE in 1h 45m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/554b218a96734a0498bb3dc36017f111 : SUCCESS in 1h 51m 39s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e2095ce4e2ee45b19845a47c8db9a318 : SUCCESS in 1h 20m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/07185843f47f48848fb31de8d3966e7c : SUCCESS in 2h 03m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/93666a2397e24701b46b0fd20d213df8 : SUCCESS in 48m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/45c2a37ae1cb47beb557ad9d888521e0 : SUCCESS in 1h 44m 38s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/59dc214afc8345f08db31f8a6f0b49a8 : SUCCESS in 2h 00m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0b8158b5339491d90def8887de9c91c : SUCCESS in 52m 01s","accounts_in_message":[],"_revision_number":3},{"id":"bf7f5055efc21428a0b31b6062120c6f76426e5f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-02 11:52:13.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/850831/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d806029/ : FAILURE in 1h 43m 25s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/850831/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/df9dfad/ : SUCCESS in 1h 41m 03s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/31/850831/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/c9fe52e/ : SUCCESS in 1h 40m 45s","accounts_in_message":[],"_revision_number":3},{"id":"4ccc8c0672e7b6bf1bfb95e708328c4417a0cdb0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-09-02 16:02:58.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/31/850831/3/check/dsvm-tempest-cisco-zonemanager-job/c11be03/ : FAILURE in 1h 05m 16s","accounts_in_message":[],"_revision_number":3},{"id":"02d29118aae20b5d9fa129682a6fd28da6cf8573","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-09-05 09:57:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"70436063c3a3faa4b139b8b1e4ee4e4699b1bb48","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 10:20:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f97efbeb26d4dbf9c41adce6f2f05ba\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ee67ec5496d46039a6f8223506a4aa3 : SUCCESS in 19m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1211c7892b0447198e7bf61d2ecdeaad : SUCCESS in 21m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1b36a1f44a42ca77d5efddf95f192b2b5a914096","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-09-05 11:04:46.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/850831/4/check/iscsi-driver-alletra-6k/e167e7e/e167e7e.html : FAILURE in 1h 05m 53s","accounts_in_message":[],"_revision_number":4},{"id":"81916a566ac400315e82ea0b65fcc49b0e8a38d3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-09-05 11:12:23.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/31/850831/4/check/dsvm-tempest-cisco-zonemanager-job/9c8ce6c/ : FAILURE in 1h 06m 04s","accounts_in_message":[],"_revision_number":4},{"id":"222d611743ee213f2645208466ac307ae08737cd","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-09-05 11:14:27.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850831/4/iscsi/ : SUCCESS in 1h 15m 51s\n- cinder-smb http://cloudbase-ci.com/cinder/850831/4/smb/ : SUCCESS in 51m 32s","accounts_in_message":[],"_revision_number":4},{"id":"ce9777b3c578f70ba214ec8e393b526e4ff8dc7a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-05 11:23:30.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850831 : SUCCESS in 1h 24m 07s","accounts_in_message":[],"_revision_number":4},{"id":"fea1770f5b4c2f14ed47f5f2e6c3adce91ca4046","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-05 11:25:53.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/850831 : SUCCESS in 1h 26m 37s","accounts_in_message":[],"_revision_number":4},{"id":"cf3839b3169cc9948b15d02f6fc9b3118235d7c3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-05 11:47:21.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/850831/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/11e9705/ : SUCCESS in 1h 32m 58s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/850831/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7a6382b/ : SUCCESS in 1h 48m 09s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/31/850831/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/69db88a/ : SUCCESS in 1h 48m 24s","accounts_in_message":[],"_revision_number":4},{"id":"37df1306fd87045aca0469cede6b1f1b3d4b1d68","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-05 11:50:42.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/850831/4/check/DellEMC_Unity_FC/1a22e63/DellEMC_Unity_FC/1686 : FAILURE in 1h 52m 39s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/850831/4/check/DellEMC_Unity_iSCSI/2bf99eb/DellEMC_Unity_iSCSI/1546 : FAILURE in 1s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"2bf22326d9071cd1a568ead4422053f7d835bb9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 12:08:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e597ec55e7e14d6099bb974dcc96d043\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b4105746786415ab8e7fa69d943eaff : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47d7635cbdec4858bc606d8fda79fb9b : SUCCESS in 11m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eed5a1cfa0734f8ab6c5743b50075923 : SUCCESS in 10m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b71ecaa0c0a40da98491ba541c1288e : SUCCESS in 11m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32d60e540677449aba1614f06095cf49 : SUCCESS in 12m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/9ace54cc4b604371b3937f99d11e424d : SUCCESS in 58m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/37202504e43848ed8b6a37e3a198da91 : SUCCESS in 1h 21m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b7abbf64aec4c57adcff92487ac562c : SUCCESS in 1h 06m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bd9c61f1b35497c8a91c388c5103392 : SUCCESS in 8m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0aba641053a94e2ab737270a378dfa5d : SUCCESS in 16m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d71c970d6f214b70886a1abef544b757 : SUCCESS in 6m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d0ee1614e230482baf90b66275156fdd : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/94ea464e055d48a0973e8ce7d5d5031b : SUCCESS in 6m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3d89d8fd163c4dc6a29493cee1519fd8 : SUCCESS in 7m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2f882c61a0944fe6b789b2f6765afff3 : SUCCESS in 34m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2abea990b88f4b3793b1d7de4af7a3cd : FAILURE in 6m 46s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fd13f333754c443cb208daf1547c51b5 : SUCCESS in 1h 11m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/14a6bcc77ea24586b719605d15287ec5 : FAILURE in 1h 26m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f05c21bd44a549168ad1e6091a0b1471 : SUCCESS in 1h 33m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/02baf1509a52471691062d149b117da3 : SUCCESS in 1h 10m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8b6a8d487d5a4bb1a923be1ef9df7fb5 : SUCCESS in 1h 20m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/29ab044409134d658aeb59d39ba2bf06 : SUCCESS in 45m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c2a8724ed0f04ac7bf3229d570778c6e : SUCCESS in 1h 21m 57s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/88b26c41d48f4b0fb2570c843ec2c65d : SUCCESS in 2h 09m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07b909ba28dd4b3d9262e1d93c56f7eb : SUCCESS in 55m 15s","accounts_in_message":[],"_revision_number":4},{"id":"b36ea36e23b956a2cd133235816617f5501db3a1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-05 13:26:39.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-850831-4/ns5_nfs : SUCCESS in 3h 28m 26s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-850831-4/ns5_iscsi : SUCCESS in 3h 27m 8s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"8fc2af72c8f7153b681de405e62033760a49bcfc","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-05 13:27:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/850831/4/upstream-check/cinder-tempest-plugin-ontap-fcp/4256040/ : SUCCESS in 1h 16m 53s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/850831/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/4caf97d/ : SUCCESS in 43m 02s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/850831/4/upstream-check/cinder-tempest-plugin-ontap-nfs/d52756d/ : SUCCESS in 1h 00m 41s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/850831/4/upstream-check/cinder-tempest-plugin-solidfire-iscsi/f37526c/ : SUCCESS in 1h 01m 43s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"692632c32a4323ef555236206ac53a3b75fef013","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-05 14:29:24.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/850831/4/check/DellEMC_Unity_FC/f6e9995/DellEMC_Unity_FC/1688 : FAILURE in 1h 20m 26s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/850831/4/check/DellEMC_Unity_iSCSI/aba1dd5/DellEMC_Unity_iSCSI/1549 : FAILURE in 1s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"3401a782c2acc1b17285b777929d2db8cfd9b8b0","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-09-05 15:09:06.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/31/850831/4/check/DellEMC_PowerStore_iSCSI/cbd534c/DellEMC_PowerStore_iSCSI/703 : FAILURE in 8s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/31/850831/4/check/DellEMC_PowerStore_FC/72da77c/DellEMC_PowerStore_FC/858 : SUCCESS in 1h 20m 15s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/31/850831/4/check/DellEMC_PowerStore_NFS/652d72a/DellEMC_PowerStore_NFS/393 : ABORTED in 3h 00m 01s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/31/850831/4/check/DellEMC_PowerStore_NVMeTCP/0b684dd/DellEMC_PowerStore_NVMeTCP/268 : FAILURE in 1h 43m 36s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"0d863308ec72c88292bc0fc85e915fba5c1fc264","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-05 17:27:26.000000000","message":"Patch Set 4:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/551","accounts_in_message":[],"_revision_number":4},{"id":"a0192d223fe5d7e5ef2b6cba33eba49342927b40","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2022-09-05 19:00:06.000000000","message":"Patch Set 4:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/850831/4/fc : SUCCESS in 1h 09m 53s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/850831/4/iscsi : FAILURE in 4h 17m 35s","accounts_in_message":[],"_revision_number":4},{"id":"75691c46bb42d743c35cd75a4bf4ecc8363ac38a","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-09-05 22:27:59.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"8eca6a2132a155453748987b3c7aeed978cb91d5","author":{"_account_id":28403,"name":"Naoki Saito","email":"nasaito@nec.com","username":"n-saito"},"date":"2022-09-06 06:20:27.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"84690e3c2f20e5066c06a00aa83293dd0f0adac4","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-09-07 00:57:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a33c7bbdd4ffa57e2b97f20c571644c302f43593","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2022-09-07 05:00:06.000000000","message":"Patch Set 4:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/850831/4/fc : SUCCESS in 1h 09m 53s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/850831/4/iscsi : SUCCESS in 1h 34m 54s","accounts_in_message":[],"_revision_number":4},{"id":"42c83b374ce2ab5a8d49218958857280b746ffcf","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-09-07 06:07:35.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/850831/4/ : SUCCESS in 1h 35m 20s","accounts_in_message":[],"_revision_number":4},{"id":"010d7e36f562115fce6908bda09415a196284a0e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-09-10 20:36:19.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/850831/4/20220911020103/NEC-FC : SUCCESS in 3h 32m 13s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/850831/4/20220911020103/NEC-ISCSI : SUCCESS in 3h 35m 13s","accounts_in_message":[],"_revision_number":4},{"id":"6e7d6cc9eddb691e1206acae26ecfa4e67121aec","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-09-13 08:33:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/755f1867343e4a98b7d1c79351308049 : SUCCESS in 42m 32s","accounts_in_message":[],"_revision_number":4},{"id":"b6dbd7b36b97f2d934b2eb255f7a50c7254267af","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-09-18 06:06:58.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/850831/4/20220918095006/NEC-V-ISCSI : SUCCESS in 2h 20m 02s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/850831/4/20220918095006/NEC-V-FC : SUCCESS in 5h 16m 49s","accounts_in_message":[],"_revision_number":4},{"id":"1520eb21fd188722804d7463e2495c1de770aaa4","author":{"_account_id":22879,"name":"Prabhuraj K","email":"k.prabhuraj@in.ibm.com","username":"pk1218"},"date":"2022-12-21 09:13:46.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"24ff14dc3f2cad52890549cf93f9557105cc2236","author":{"_account_id":14806,"name":"Chhavi Agarwal","email":"chhagarw@in.ibm.com","username":"chhagarw"},"date":"2023-01-03 08:50:05.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"d5f1dd41100497064545e3d785fafc111122f51c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-02-09 07:55:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1d787fb306e7a5c5eaae11b695c3492f3ccb47e7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-09 08:00:56.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"db6246542a86b1999a4db8701d4e32d6f17d8089","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-09 08:01:54.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/8c48af892312443b84b20e89de57fd10 : POST_FAILURE in 5m 23s","accounts_in_message":[],"_revision_number":5},{"id":"5e69934cea09b432a1b822576a9fe62c8a5d3f74","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2023-02-09 08:15:58.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/31/850831/5/check/dsvm-tempest-zadara-driver/f3fb816 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 19m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d039892bca81c4416d652348dfb9c3fc25e8f963","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 08:17:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed2bb5c394cb4845bb0d653fa469dd48\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b486dfeac5a4fd3ae0678ad8519e0ad : SUCCESS in 19m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a199d31dbdd64fcaa694880b88131ca7 : SUCCESS in 19m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cd70504f63df0ee75379f41439f20fde76a64de3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-09 08:20:54.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/850831 : FAILURE in 23m 34s","accounts_in_message":[],"_revision_number":5},{"id":"89a481c9365ad5749da79af8c6d2e8af86b87cd3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-09 08:23:03.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/850831/5/check/iscsi-driver-alletra-6k/8898f72/8898f72.html : FAILURE in 26m 05s","accounts_in_message":[],"_revision_number":5},{"id":"884fa3732a1e90746e6f891a33847099b0015141","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-09 08:24:15.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850831 : FAILURE in 26m 24s","accounts_in_message":[],"_revision_number":5},{"id":"d76c65d37cac15a7820d8963bcc2632a8cd1b076","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-09 09:13:40.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ae33a8af8fe74e1b8ad5c0049f56aef1 : SUCCESS in 1h 15m 45s","accounts_in_message":[],"_revision_number":5},{"id":"7119dcdb8128bfafa36ccebe928bea1f2f066f11","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-09 09:33:10.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/850831/5/ : FAILURE in 1h 01m 54s","accounts_in_message":[],"_revision_number":5},{"id":"4f208bd0c0db3d50a72da1d8f1884d5e4a0ab19e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 10:34:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c26fde752de4cd792f8ecd576a33dbe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5aa612ce7fc3419f9ce909bee8b702db : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e7762feed86480c858a0c4392b80a65 : SUCCESS in 11m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10d8b2b84ef74659aee5cc6d7fee8911 : SUCCESS in 11m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47284b34e8aa4fc7ae8fc223f8e52bf7 : SUCCESS in 11m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/a26ce5e7acfb41adb003b06d21f6032c : SUCCESS in 50m 14s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c501eca418644455901fb8049bb1cc96 : SUCCESS in 55m 25s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cf86b440c14846158c68fa62d6207685 : SUCCESS in 1h 29m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53d95eade821494babaec587569d9c61 : SUCCESS in 53m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d559f7305ee41afbf816a9a92f62ef0 : SUCCESS in 8m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fa13d735c8e44299acc7f869d1be209f : SUCCESS in 14m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/62d7c3b3a0dd4cf18ac2cca43303266b : SUCCESS in 5m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2cf2848e46c3489c87d5618ea99bb52b : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3a9efb0bd1f841a0a5a6e754340a050d : SUCCESS in 6m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4f16c32d4f4f40eda30bec42e70e38df : SUCCESS in 6m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ae00cc9a4001473e96aff5e35220b23b : SUCCESS in 58m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/387545792a6f4d69ac1150e4c8d8f930 : FAILURE in 7m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6492e830a0ba437aac51a8312a28e330 : SUCCESS in 1h 29m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b50b1e9cb5a043d39189a8f23a5fb3db : FAILURE in 1h 47m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/68c0debe5a7a42a8adc591f4965bc14c : SUCCESS in 2h 19m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d8e0008d90874bbf84564f4884ef73ad : SUCCESS in 1h 27m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cb077a55d36041c8ae3d6070824a7b08 : SUCCESS in 1h 43m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f83a639eb6394fc1a3efc986ccb7e75c : SUCCESS in 46m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/47040ac129c3423798d9d8fc8e127c54 : SUCCESS in 1h 22m 39s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/50f314f1ff894089b7d96ab5d8f0019b : SUCCESS in 2h 30m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4aa4e8ed3e43418f9ba0130e31fd1ce6 : SUCCESS in 41m 46s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/302ba452ded8488bad5ff1b9a52f543b : SUCCESS in 1h 21m 47s","accounts_in_message":[],"_revision_number":5},{"id":"a9c275b07536ca447721f80588561bc927b7b2e3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-09 10:41:35.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/850831/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/de95b16/ : SUCCESS in 1h 40m 31s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/850831/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/49b0b32/ : SUCCESS in 1h 36m 35s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/31/850831/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/05d3b31/ : SUCCESS in 1h 39m 47s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/31/850831/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/cc2aa28/ : SUCCESS in 1h 41m 59s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/31/850831/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/412d22a/ : FAILURE in 2h 44m 44s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/31/850831/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/1b20d62/ : SUCCESS in 1h 49m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5ac1078f5b80f7cef28d216c1e270749e8b299fe","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-02-09 15:53:08.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/850831/5/20230210002436/NEC-V-FC : SUCCESS in 27m 55s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/850831/5/20230210002436/NEC-V-ISCSI : SUCCESS in 28m 29s","accounts_in_message":[],"_revision_number":5},{"id":"c56069183a331b43a4ca8b44aa6a9656298c0876","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-02-09 17:05:04.000000000","message":"Patch Set 5:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/31/850831/5/fc : SUCCESS in 2h 07m 37s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/31/850831/5/iscsi : SUCCESS in 2h 27m 15s","accounts_in_message":[],"_revision_number":5},{"id":"6f8f10fa950d83def7fccf9a0b809683210dfcf1","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-02-09 18:47:46.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/850831-5/HPMSA-iSCSI/ : SUCCESS in 1:38:33\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/850831-5/HPMSA-FC/ : SUCCESS in 1:17:09","accounts_in_message":[],"_revision_number":5},{"id":"2d1473caaf55cfa256d39d70046f6cb33e1dec56","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-09 21:30:09.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"b1cc3b084552e36a870e64f7dcc102acd7922d9a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-09 21:31:43.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-850831-5/ns5_nfs : FAILURE in 9h 27m 15s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-850831-5/ns5_iscsi : FAILURE in 13h 35m 20s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"8f9c7b6fcfcf6c0f3b2e6c9206b6f75a1c0cf39e","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-02-09 23:22:37.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/850831-5/Seagate-iSCSI/ : SUCCESS in 1:48:54\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/850831-5/Seagate-FC/ : SUCCESS in 1:31:47","accounts_in_message":[],"_revision_number":5},{"id":"c479eae9da4a3f8ea6b5778733caaa922db3872a","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-02-10 05:29:58.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/850831-5/PVME-iSCSI/ : SUCCESS in 1:50:30\n- PVME-FC https://os-logs.seagate.com/dell-ci/850831-5/PVME-FC/ : SUCCESS in 1:32:19","accounts_in_message":[],"_revision_number":5},{"id":"4a6a8952bcb6812807b7444d09466d371386d2b8","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-02-15 21:00:05.000000000","message":"Patch Set 5:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/850831/5/fc : SUCCESS in 0h 24m 23s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/850831/5/iscsi : SUCCESS in 0h 50m 03s","accounts_in_message":[],"_revision_number":5},{"id":"4cbc92211326132c546e540a6a20d015ce3c448e","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-02-17 14:49:20.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"972023a9048ffc2cb0fcb4f143141123c331974d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-02-17 14:51:41.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9165553217d5e1fc46e7c20cafcbdd5e4340fc44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-02-20 15:22:05.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6eece43ea100e379de144bfbb5f83054781602ba","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-02-20 15:29:15.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"cb99753f64f2f8497d181dbaa46cb8277d18ca20","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 15:32:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12d21dc0517f426f9be4cde91943d185\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e58917b15024c31bc62ff73692a36c8 : FAILURE in 9m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b9ab7da36c9b41b0bf731ba1dc890133 : FAILURE in 8m 34s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b257588f2f4e90cf008c024a9f09ad7cafb17593","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-20 15:55:36.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/850831 : FAILURE in 23m 24s","accounts_in_message":[],"_revision_number":6},{"id":"5a9532ecb5fdab2c787e74980bbb7e5a94ca3f2c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-20 15:58:45.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/850831/6/check/iscsi-driver-alletra-6k/f72895c/f72895c.html : FAILURE in 26m 17s","accounts_in_message":[],"_revision_number":6},{"id":"ee07ab21e81c96e31ba3d1796d6c348cb05ab0af","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-20 15:59:22.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850831 : FAILURE in 26m 10s","accounts_in_message":[],"_revision_number":6},{"id":"7b4d1f79f9d5dbe9c08d668b751bb8640251c150","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-20 16:32:39.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/02d2ac462e2042438623eee55194beb2 : POST_FAILURE in 29m 30s","accounts_in_message":[],"_revision_number":6},{"id":"53b2c2ae7c80ba1b4558148985003307c70e9a69","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-02-20 16:51:44.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/31/850831/6/check/fujitsu-eternus-dx-iscsi/e0dfd98 : SUCCESS in 1h 29m 02s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/31/850831/6/check/fujitsu-eternus-dx-fc/10fce3c : FAILURE in 3m 14s","accounts_in_message":[],"_revision_number":6},{"id":"2df0c789a544887d19e00cd8db676d655a84043c","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-02-20 17:12:29.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8b272098aa070694830044a1baa6295be76f6252","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-20 17:19:42.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/850831/6/ : FAILURE in 1h 03m 25s","accounts_in_message":[],"_revision_number":6},{"id":"6354d451aba5fbc77507d01ebe0d04859096572b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-20 17:41:30.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/850831/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/6ae058d/ : SUCCESS in 1h 43m 57s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/850831/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/61f3e2e/ : FAILURE in 2h 01m 51s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/31/850831/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/e87e3b0/ : SUCCESS in 1h 40m 48s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/31/850831/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/faeb587/ : SUCCESS in 2h 08m 08s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/31/850831/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/0306fc0/ : SUCCESS in 1h 51m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1374e08cec1d7b8c3ad3c6157d870294b185ba75","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-20 18:37:27.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/850831/6/check/DellEMC_Unity_FC/d305ee2/DellEMC_Unity_FC/2328 : SUCCESS in 1h 19m 52s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/850831/6/check/DellEMC_Unity_iSCSI/be243ef/DellEMC_Unity_iSCSI/2182 : SUCCESS in 1h 42m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"90388eefae4185299366e76b84ab13b783f0c3e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 19:00:05.000000000","message":"Patch Set 6: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba62947874474e1ca583a5474513ac72\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/962880da42f74dd08ede263334a6c348 : FAILURE in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27a0b43c982f4c518419564e8ea57fd3 : FAILURE in 11m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d722157edb6f4119b7e4b3cffd1fa716 : FAILURE in 12m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b183fabaa09546379af0ee4fae9051e1 : SUCCESS in 14m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/fb9b5626e3f14465bf8f053b0540d791 : SUCCESS in 53m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/45dc915df7e9479d871eb4564c131a72 : SUCCESS in 52m 14s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/233aa8e997184b88990f4dfbffe61dda : SUCCESS in 1h 13m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/edf8b3337c6e4c4faee02616a07d6334 : SUCCESS in 49m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fc355edc63f472f82a5ac20b9dfe23c : SUCCESS in 8m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a9896e9874c9426cb0227b6106b9231d : FAILURE in 14m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/98a88f723b7245aea1c2567f110c159e : SUCCESS in 5m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/75fde6cd604c423cb6fd6c2e799250f5 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d86561acdf884fada248095af586901d : SUCCESS in 5m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4126794af4ca4e03bc93e8dda081eee0 : SUCCESS in 5m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/345eae472f39461bb4f657655c16f780 : SUCCESS in 50m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3989a8cf46804fe4a813e79b50e566b9 : FAILURE in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a55e6d929fc94e6887b27347abdcbcd6 : TIMED_OUT in 2h 05m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/281fe7ebbc3042a39357e834b8075af9 : FAILURE in 1h 42m 06s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1bb862f10f5f42a398bc44deeb5491ad : SUCCESS in 1h 44m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ff5daf83716b40509466434b0d178c38 : SUCCESS in 1h 06m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1b1bb5252a6f457b891b95911f94d6bd : SUCCESS in 1h 33m 50s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d80e2d06f27a44cbaeb87df71933aaef : SUCCESS in 44m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/79f170f92dcf4bb7b6ffeefbb8fec731 : SUCCESS in 1h 26m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/37d5231fad02455ba14b79df1852c30a : SUCCESS in 2h 25m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09f4a8c680884fbb98ca79fd7316fc14 : SUCCESS in 46m 40s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2060d6982c2b41339aa9184b29f5eaaf : SUCCESS in 1h 14m 55s","accounts_in_message":[],"_revision_number":6},{"id":"6f89127a54511e10d9f3186c7858374db01f2b5f","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-02-20 21:00:06.000000000","message":"Patch Set 6:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/850831/6/fc : SUCCESS in 0h 51m 32s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/850831/6/iscsi : SUCCESS in 1h 47m 04s","accounts_in_message":[],"_revision_number":6},{"id":"dbfddac460ae7265beafe0aa48328774932afa52","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-02-20 21:10:19.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/850831/6/20230221003258/NEC-V-ISCSI : SUCCESS in 2h 07m 03s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/850831/6/20230221003258/NEC-V-FC : SUCCESS in 5h 37m 17s","accounts_in_message":[],"_revision_number":6},{"id":"c6bccb418e691b238143b460d6a0013395461297","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-20 23:02:58.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-850831-6/ns5_nfs : FAILURE in 7h 3m 6s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-850831-6/ns5_iscsi : FAILURE in 7h 40m 23s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"78fe2ad34e67779af49764a5f58e2613dfa81dd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-02-21 01:46:32.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"24e1907b7f4c02446ad9e2eb18939b0a82a51d4a","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-02-21 01:59:59.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a5a900478c384f1bb90b48d8d0e22cfc89d5dc57","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-21 02:11:32.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/850831 : FAILURE in 23m 39s","accounts_in_message":[],"_revision_number":7},{"id":"afe445c373e3e4463f496310f35de4819cfee380","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-21 02:12:51.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/850831/7/check/iscsi-driver-alletra-6k/3c30d1b/3c30d1b.html : FAILURE in 25m 18s","accounts_in_message":[],"_revision_number":7},{"id":"226526ade4b37436fbf6dce1b3225bb4eefd02df","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-21 02:14:14.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850831 : FAILURE in 25m 57s","accounts_in_message":[],"_revision_number":7},{"id":"802b50657af531dc4be9853f0757dd0e6973f209","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 02:15:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a06e3312f374675ad5e36f1d8a2e7d2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dc917f826af84aa6a3396f15ca8c96a3 : FAILURE in 22m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4b67ae489f524fddbb41518b0d49e916 : FAILURE in 26m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"08e4cb4e3e770adaaf3a7afc0b5e58e9ff59e720","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-21 02:54:09.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/850831/7/ : FAILURE in 1h 00m 01s","accounts_in_message":[],"_revision_number":7},{"id":"06971b83f9b06132106e5bb5070a8e1f9fa033a6","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-21 03:12:24.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/c30686c6e4e840e2bc3e0f06e9fb6c5e : SUCCESS in 1h 25m 10s","accounts_in_message":[],"_revision_number":7},{"id":"b181eef84f136423c60f1d717c085c027493da0b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-21 03:45:21.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/850831/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2c6209f/ : SUCCESS in 1h 45m 17s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/850831/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/653cba0/ : FAILURE in 1h 55m 57s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/31/850831/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/16fed39/ : SUCCESS in 1h 36m 44s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/31/850831/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/361fbcd/ : FAILURE in 1h 55m 13s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/31/850831/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/5691f97/ : SUCCESS in 1h 47m 21s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c34bfcd510d6e41bf26377efb0b8a2a5c0759abf","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-02-21 03:54:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"51341f6b415840405f9c86892cb4ccd5201b846c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-02-21 04:05:03.000000000","message":"Patch Set 7:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/31/850831/7/check/fujitsu-eternus-dx-iscsi/5f03e1a : SUCCESS in 2h 17m 53s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/31/850831/7/check/fujitsu-eternus-dx-fc/3fc8489 : SUCCESS in 2h 13m 03s","accounts_in_message":[],"_revision_number":7},{"id":"57bacb042f3aa52cc155cb0c467e06b59a30e3a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 04:20:32.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a96a600a0714a7db623c5ceb34ffc83\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/890b0fe685ae4b9a9a24c427da269a29 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a571ef2dcd2b480c96f22fa77424ce50 : FAILURE in 10m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5ddba1b4501405e978cd848bffb80e7 : FAILURE in 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4397abaa23ac41f6b886c6f1347a1302 : SUCCESS in 13m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/6dcbf0030ea640adb2a60436f30d5695 : SUCCESS in 56m 48s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6b84d6f4039a4374a3da98adbfbc55e1 : SUCCESS in 1h 01m 50s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/64f2f21fc9b44d4880ff26a104e74da6 : SUCCESS in 1h 18m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8f43b8fb5884cf7b21c067617a2611a : SUCCESS in 55m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b5bfb25386f4543b0b3d9a0bdf4f31f : SUCCESS in 7m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d1e3822cbbc2442d9cc67124ca99ebd7 : FAILURE in 15m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/05527f62c3df4c0a97064ba38e20b635 : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0609e9dad5644f819849eeaf07591a77 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5bee21d045294fcf973849a369d5374c : SUCCESS in 6m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ad451d7bc5344fc9ba08957a018e1e5f : SUCCESS in 6m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/568edda7a7d64c24ab1922da24f6a74e : SUCCESS in 35m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e58b52316a5840669c284018fdde17bc : FAILURE in 7m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2f7cad8f1ce44ccfb58bb08c8e063a21 : SUCCESS in 1h 12m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e764d1a501874574bfddd75779ad2bbd : FAILURE in 2h 31m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a794f719fb964438b83ef310ca7f64c1 : SUCCESS in 1h 44m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b191de72d97e425ea56afc3d81b90f58 : SUCCESS in 1h 09m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9176393c21df4270ae6b7a9208db115d : SUCCESS in 1h 34m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f55db34c3e1b41f5a5a02f8e524f94d3 : SUCCESS in 28m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2160f1626af141d7ab9f40feb29c3310 : SUCCESS in 1h 21m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/588763f8ba28443a9da5aa66758b60f7 : FAILURE in 2h 28m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e96d4e98b7ac40f3b49eea211d2fb7d8 : SUCCESS in 50m 58s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e74ac2a2f4e24852b20d8ece88613f04 : SUCCESS in 1h 33m 03s","accounts_in_message":[],"_revision_number":7},{"id":"b7c3e5fb772279a57b057b3ecb1e130728b42e33","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-21 04:24:11.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/850831/7/check/DellEMC_Unity_FC/eefb7ff/DellEMC_Unity_FC/2329 : SUCCESS in 1h 24m 29s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/850831/7/check/DellEMC_Unity_iSCSI/2b70aa6/DellEMC_Unity_iSCSI/2183 : FAILURE in 2h 37m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"e2b25e11622bbbe3c2e6b410927fea050d3b95f2","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-02-21 06:33:50.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/850831/7/20230221105447/NEC-V-FC : SUCCESS in 2h 43m 13s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/850831/7/20230221105447/NEC-V-ISCSI : SUCCESS in 4h 38m 59s","accounts_in_message":[],"_revision_number":7},{"id":"d46e0f0aacd4423c3858128bd6cdf324e43972f9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-21 08:13:38.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-850831-7/ns5_nfs : FAILURE in 5h 42m 54s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-850831-7/ns5_iscsi : FAILURE in 6h 26m 35s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"4cab5cd47e3b559acf46de2c0e71c1b7c90221b3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-21 09:06:52.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/ea2829eea311437dad1820a6049acb70 : SUCCESS in 2h 17m 45s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/384aee8cb9be4647a65bb7dd6ab4f6a5 : FAILURE in 1h 50m 39s","accounts_in_message":[],"_revision_number":7},{"id":"41eaa9602b0bd66ffaa8eee185ed44d544140e84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-02-21 11:56:05.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"860c80e83a293c27d5bea2399fb8a367ad569519","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-02-21 12:00:47.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"11c165b37a7b5538504f9eb763a30db1ef3d7f76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 12:05:52.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa4364a5151745e194a762d57465424e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c46f13a7f3649b8a8d08ddcf2966fdb : FAILURE in 8m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/19f6855fca454811836bb7757ff31ebd : FAILURE in 8m 33s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"29b98689b7cf07c7c97830ee7ca68b4719e37ee2","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-02-21 12:18:59.000000000","message":"Patch Set 8:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/31/850831/8/check/fujitsu-eternus-dx-iscsi/d514cd5 : FAILURE in 22m 18s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/31/850831/8/check/fujitsu-eternus-dx-fc/bf74431 : FAILURE in 21m 27s","accounts_in_message":[],"_revision_number":8},{"id":"f51c126bd9add45ed7eb87a417b13d7a423c0624","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-21 12:24:51.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/850831 : FAILURE in 23m 32s","accounts_in_message":[],"_revision_number":8},{"id":"6b0ac09551def795efa12548646bad33edb0e80f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-21 12:25:14.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/850831/8/check/iscsi-driver-alletra-6k/83dcb9e/83dcb9e.html : FAILURE in 25m 59s","accounts_in_message":[],"_revision_number":8},{"id":"1b855796c22111714d9cee1158729a170c343854","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-21 12:27:25.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850831 : FAILURE in 26m 31s","accounts_in_message":[],"_revision_number":8},{"id":"47056da58d9ca39a0ee0aabd2bf3f67fa4c0d781","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-02-21 12:52:37.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"df30cb78b29bafa1952196c505290bbf7f93017f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-21 13:14:22.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/850831/8/ : FAILURE in 1h 10m 32s","accounts_in_message":[],"_revision_number":8},{"id":"1230ccc9840e3a933a3f635f60ec948f11e05e3c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-21 13:44:31.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-850831-8/ns5_nfs : FAILURE in 1h 4m 27s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-850831-8/ns5_iscsi : FAILURE in 1h 3m 26s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"6eb85d6f8e4ed6f025440e113552bbe79516f734","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-21 13:57:55.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/c2df021d919841eea3735c0aa5fdc0f2 : TIMED_OUT in 2h 01m 12s","accounts_in_message":[],"_revision_number":8},{"id":"7bcf59a13a618a1362d7b9ed29edcda4682b10f3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-21 14:07:10.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/850831/8/check/DellEMC_Unity_FC/e4819e3/DellEMC_Unity_FC/2330 : SUCCESS in 1h 30m 39s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/850831/8/check/DellEMC_Unity_iSCSI/d471911/DellEMC_Unity_iSCSI/2184 : FAILURE in 2h 10m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"cbceb3068097bbe64f16e735b75381240485530b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 14:19:33.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d552bc547a694ac5b66239d9c121f8ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e2ec3e0efb744df9f7bb186c7c5fa94 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9d14e28b9614d98a45a044bef5ad183 : FAILURE in 10m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2c9426bcccc4b38a3372b7542644c1f : FAILURE in 11m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02c1e79034e7469c81387c8d9c843203 : SUCCESS in 10m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/057496d735584dc18f2b796a9633ba94 : SUCCESS in 45m 27s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4d3c0677176d4673b895d35410334d49 : SUCCESS in 44m 51s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3ed41019b9814ef296ae5da86ddeb187 : SUCCESS in 1h 28m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef1a3631bdda4ae8a7138a68ee049585 : SUCCESS in 46m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e05bc76ea991433e922a83f9fee1e7a0 : SUCCESS in 6m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4c3918e98c4347f0bc0a01217deb4918 : FAILURE in 13m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/07d23e7e89a146c1b72ab60e92f85dc1 : SUCCESS in 7m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e4ad82aa73844b3f914e57bdc694601c : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/341be7198fbd47b986cf81ebf936b10d : SUCCESS in 6m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/60a59a4275b543aab0f9403b71708c2f : SUCCESS in 5m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f91cc6c1331741189e18fa5cc01762c8 : SUCCESS in 44m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9946859e37564cffba5c4021eac4919a : FAILURE in 5m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5af0c443007a49d68d6910bae7b9e619 : SUCCESS in 1h 59m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ab7ecf151c0f419fbf354c4ac6a88a31 : FAILURE in 1h 28m 18s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0b80ba88cf01491a8a3f965d73dd28b5 : SUCCESS in 2h 12m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e8f338f8482443398339f94ed53ce374 : SUCCESS in 1h 16m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3101d273505c4a9d8120837f32f33363 : SUCCESS in 1h 24m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8cdda9c1f4b1420c85296b3ab7051f02 : SUCCESS in 40m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/19c30917c2a94a02be8b8fc77ac0c48c : SUCCESS in 1h 21m 16s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e2867f0e3bc44ac095038c602e96de1b : FAILURE in 2h 19m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6360130006042639239ca0304cc112f : SUCCESS in 57m 51s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46338c3ff29449619ac3fe2aa0781fc4 : SUCCESS in 1h 16m 45s","accounts_in_message":[],"_revision_number":8},{"id":"241ce692299284b09c2341dbc8679d64325289d0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-21 14:41:05.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/850831/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b3a732e/ : SUCCESS in 1h 54m 56s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/850831/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/665cc0f/ : FAILURE in 2h 42m 42s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/31/850831/8/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/7e1d10b/ : SUCCESS in 1h 44m 44s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/31/850831/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/cdbfa1f/ : SUCCESS in 2h 02m 56s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/31/850831/8/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/b7ce7a9/ : SUCCESS in 2h 33m 43s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7199098ba9615fa5dacb401d465840a381dabd49","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-21 15:32:06.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/f3fbeb7180304c468075ea9f854846cb : SUCCESS in 2h 13m 29s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/8cd5ee44759c4cee9aba90f63c383cf0 : FAILURE in 1h 44m 49s","accounts_in_message":[],"_revision_number":8},{"id":"0451d1ab7be7d1a09d2279cae6b4544ea426ca7a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-02-21 19:03:54.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/850831/8/20230222013557/NEC-V-ISCSI : SUCCESS in 1h 57m 45s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/850831/8/20230222013557/NEC-V-FC : SUCCESS in 2h 27m 54s","accounts_in_message":[],"_revision_number":8},{"id":"44580ba1f6fb5d5fea150205d2c68ccd0e9cdd06","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-21 22:43:23.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/528475aed0cc41a9ab60860f026874a0 : SUCCESS in 1h 21m 37s","accounts_in_message":[],"_revision_number":8},{"id":"fd1ca5ed79e382886b8a6b5dcd88b0a23fdb853b","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-02-22 06:17:29.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4428d484f2037bed3dfa82407acacd8e515ed337","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-02-22 06:20:42.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"846325991803cbd1f9b3d03ebd1490f204c24944","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 06:30:31.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3a1c2e855524d87a620e8808a405fd5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa43a8097f314d79ab60f6c6bec1b394 : FAILURE in 8m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3c795ba4b1994e268ce7415dd17bfcb7 : FAILURE in 8m 07s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f64aea46ff664ca480eb29fca03f66bc7a4668dc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-22 06:45:40.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/850831 : FAILURE in 23m 30s","accounts_in_message":[],"_revision_number":9},{"id":"fb9f89b91a63544c2a878f25e267ec427f9e4fa7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-22 06:48:00.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/850831/9/check/iscsi-driver-alletra-6k/cd71a2e/cd71a2e.html : FAILURE in 26m 06s","accounts_in_message":[],"_revision_number":9},{"id":"5c8d3fbdbc320a68bded3e942088624e39beb882","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-22 06:48:21.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850831 : FAILURE in 26m 06s","accounts_in_message":[],"_revision_number":9},{"id":"999d8495ceff9945f98a4acc24a4bef18bd451a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-02-22 06:51:26.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c0c3fc162418e9342f4a79374c10485a50d125df","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-22 07:17:56.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/850831 : FAILURE in 23m 40s","accounts_in_message":[],"_revision_number":10},{"id":"d89d98c406b71dc06d3b888047d8e158f8215133","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 07:19:35.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d03c7fe3f83742e9bf631d6817e4012a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e179bb4693ac4e28be3425e9bc73b43d : SUCCESS in 21m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4c6c6a0fec5948189cc075baef1a48ff : SUCCESS in 26m 03s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"772b3c08d39a01e2e7809004333c03bbcd3cef74","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-22 07:20:19.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/850831/10/check/iscsi-driver-alletra-6k/b164d16/b164d16.html : FAILURE in 26m 50s","accounts_in_message":[],"_revision_number":10},{"id":"74b374010dcb19441122c0d838c7b6c1b9dfcacf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-22 07:20:52.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850831 : FAILURE in 26m 14s","accounts_in_message":[],"_revision_number":10},{"id":"d34fe33c8de8b4ba65897c74e98a0c2d3de5828f","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-02-22 07:43:21.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"3f53177ed93519ac894f4607c29cf031c521c616","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-22 08:00:51.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/850831/10/ : FAILURE in 1h 01m 10s","accounts_in_message":[],"_revision_number":10},{"id":"5eeb06d93c834a744e142c9fc295857d196d9755","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-22 08:11:28.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/52bb31928c8345808094964e03bbe82b : SUCCESS in 1h 18m 11s","accounts_in_message":[],"_revision_number":10},{"id":"842c9a801a6790b529a8034708c075cd1cdb0181","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-02-22 08:57:06.000000000","message":"Patch Set 10:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/31/850831/10/check/fujitsu-eternus-dx-iscsi/4cf60bb : SUCCESS in 1h 58m 47s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/31/850831/10/check/fujitsu-eternus-dx-fc/e061018 : SUCCESS in 1h 57m 58s","accounts_in_message":[],"_revision_number":10},{"id":"87f7051b6bc0230dded74c34c3723f43ff4642d0","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-02-22 09:04:47.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e08da0415a38844446fbf1652145a238abce9d99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 09:33:23.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bba46bfec54c498e8e1c2a8b63a9e322\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee26c9506e184de08b08ee7d2c89dd39 : SUCCESS in 8m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f39887ec080c47b3a534c74d0b391d1a : SUCCESS in 11m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d71b8f195c164cbba4b8e02394fb7da8 : SUCCESS in 12m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2d2bb36e2b749ebbeae38afc7a74441 : SUCCESS in 11m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/69a7e4ec42b0401995f7da41e23a6391 : SUCCESS in 1h 02m 17s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d3e3eb0c855b40e182a2aa71c34abee6 : SUCCESS in 54m 28s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a0f5d5d2712848dfbfd3bf945673fd93 : SUCCESS in 1h 35m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/186a21dd3a74417c80301b41086ecd8b : SUCCESS in 1h 03m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3db8c977446248c489dc46066361a1a2 : SUCCESS in 8m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/95e313c425464b3aa89ae50ab7eba869 : SUCCESS in 16m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/94d92b73635e4bbeacf4c32c52756e2e : SUCCESS in 6m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/18d5fcdb478d42cb807e00559009caed : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f84e9db002c94829812a8ded7f565804 : SUCCESS in 5m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/372012d5550c465aade466087599ae44 : SUCCESS in 5m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/15b152c921fd425ca8d1b834f24cee19 : SUCCESS in 53m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/616c0e9073544284afbc251821ee84ce : FAILURE in 7m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5407f71f269f4324b5c430859c51b1ef : SUCCESS in 1h 41m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8563403414844a4685e0120cca7a17fb : FAILURE in 1h 18m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9ec7fff034c64a33b5068f38dfb6fde7 : SUCCESS in 2h 40m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5911fc042f88486f96f248b5470a19fc : SUCCESS in 1h 04m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/775d9dc4f4e846efa88e85d897842b0e : SUCCESS in 1h 37m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/59d1bb4b01d94cc8badad8b2a038d914 : SUCCESS in 37m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/34ec19b4fd1a4ff99561932c67ef09a6 : SUCCESS in 1h 23m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bedfd5f0a9cd44e6ab8c198dce88754c : FAILURE in 2h 12m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1085c01cde5a4dc8914232ca3c001635 : SUCCESS in 1h 19m 48s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/445462bd1a3f423cb5c6855d82bb9421 : SUCCESS in 1h 16m 43s","accounts_in_message":[],"_revision_number":10},{"id":"835af4c59a68220e1e6049cd8bd38e6960a1e365","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-22 09:34:42.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/850831/10/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/58b15bf/ : SUCCESS in 2h 02m 53s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/850831/10/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6079f5a/ : FAILURE in 2h 06m 37s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/31/850831/10/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/1fdd488/ : SUCCESS in 1h 50m 43s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/31/850831/10/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/7756198/ : SUCCESS in 1h 50m 02s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b4efea0ccc7848cc4a95c04196b10a38347acdb8","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-02-22 10:03:01.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"91a680e36713af128f1c510a0b2bd85da1fcf388","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 10:14:15.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbe8f2b1d5cc423f99a2892eb0ca4641\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0164848cf2541da8d826fb053c3909e : SUCCESS in 10m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d208c167eac04fd7be64da14299d47a5 : SUCCESS in 8m 31s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ca97f13ea29fe896112f16b39b41799597c9c283","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-22 10:20:03.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/aad2734720874a93aa196e111ec8501d : SUCCESS in 1h 35m 33s","accounts_in_message":[],"_revision_number":10},{"id":"4d76e3aa5a4f18d7d81820319bd255d30b0df551","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-22 11:49:45.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/4569c9f17b4a4980888a47df241f3f72 : SUCCESS in 1h 16m 59s","accounts_in_message":[],"_revision_number":10},{"id":"12a7e5dd217adabb430c6094f0ec34563ac008df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 12:09:11.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d667c6f1c374bc9a2b7a18c59e0acbf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97f0d98d03fb47bfb52a1b7c56669527 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f5d770b8a764cd8a82436560773b791 : SUCCESS in 11m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b7c452117b44a9492b19575ea4c7d05 : SUCCESS in 11m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/830551db4f15478d8452f978b9da8459 : SUCCESS in 12m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/871a933a4be94550a6888ca8ee4b4829 : SUCCESS in 41m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2cf0e2e5a7f14037a300825476514149 : SUCCESS in 46m 35s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5b008792642d489b9f2f92c3fd55206a : SUCCESS in 1h 19m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90469b08f7bf4799803dad603d933201 : SUCCESS in 45m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9eccc75276c343b797f60f55602a5c65 : SUCCESS in 7m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1eceb8a19b734456a1dac1fab3fb540c : SUCCESS in 16m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f2c35cb2947e4e52974625f7e9a41ccc : SUCCESS in 4m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4304265fce704e9686b92a22218b0f03 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1df54dc6104e4f72b411a264f9b51eb1 : SUCCESS in 6m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b084f995849b4e0b93b5cdc974582a26 : SUCCESS in 5m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2823e146239840cd94baa81f30216015 : SUCCESS in 49m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ac6d8063d5d941cbb9c5e1a7f3e5c563 : FAILURE in 5m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e2b5062f3bb44b70af52ad060e448073 : SUCCESS in 1h 17m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/85b3414612634551afef649ad271db5f : FAILURE in 1h 17m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b6997e260e4645a1a2c44b1c9d1af1c0 : SUCCESS in 1h 55m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/23062c9ddc3442178244b55dc885bc82 : SUCCESS in 1h 08m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/08484dd3f43c4a5aa5e5ff09b568e58a : SUCCESS in 1h 34m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3014f1b638814720aa2bc14b6571dd44 : SUCCESS in 33m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/93e15521446e41f48163102c07b39248 : SUCCESS in 1h 12m 38s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2de4f2afddb24070b569a1a7ce0a9510 : SUCCESS in 2h 04m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b8142f3ac8c414895aaec2e9abe499a : SUCCESS in 53m 44s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4840b29bc0d468bb4021f2c3a185c95 : SUCCESS in 1h 19m 11s","accounts_in_message":[],"_revision_number":10},{"id":"2de95ee795fc28d7ca5aa3b712175195fc66f42a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-02-22 12:18:22.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/22_02_2023_12_09_19\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"202f4c1eb0f7460b95b2697a65d3216c789046c1","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-02-22 12:56:23.000000000","message":"Patch Set 10: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"1388cd890df6e5a23601cb4fa36bf969f55fac7d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-22 13:18:14.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/850831/10/upstream-check/cinder-tempest-plugin-ontap-fcp/3b020e9/ : SUCCESS in 1h 08m 08s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/850831/10/upstream-check/cinder-tempest-plugin-ontap-iscsi/ab5c819/ : SUCCESS in 43m 24s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/850831/10/upstream-check/cinder-tempest-plugin-ontap-nfs/7bcde19/ : SUCCESS in 45m 29s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/850831/10/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/29aafde/ : FAILURE in 45m 13s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"ed644e6c78db0bd92686453928f1bce895ef78f9","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-02-22 13:39:23.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/31/850831/10/check/DellEMC_PowerStore_iSCSI/be1163a/DellEMC_PowerStore_iSCSI/1037 : SUCCESS in 1h 17m 52s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/31/850831/10/check/DellEMC_PowerStore_FC/8d4459e/DellEMC_PowerStore_FC/1172 : SUCCESS in 1h 29m 51s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/31/850831/10/check/DellEMC_PowerStore_NFS/0c0ac24/DellEMC_PowerStore_NFS/676 : SUCCESS in 1h 21m 05s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/31/850831/10/check/DellEMC_PowerStore_NVMeTCP/105325e/DellEMC_PowerStore_NVMeTCP/541 : SUCCESS in 1h 25m 47s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"aed483c8ee2a804e5fea1b12b9c7f36a74e22cc0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-02-22 13:44:36.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/31/850831/10/check/DellEMC_PowerFlex/8d654b4/DellEMC_PowerFlex/1638 : SUCCESS in 1h 35m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"92f0ac30396f75c5f25ad59a814da5214d2f1225","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-22 14:07:36.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/850831/10/check/DellEMC_Unity_FC/5fc2553/DellEMC_Unity_FC/2343 : SUCCESS in 1h 18m 18s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/850831/10/check/DellEMC_Unity_iSCSI/7511684/DellEMC_Unity_iSCSI/2197 : FAILURE in 1h 07m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"262adff08fa91d02f6181e3fe6844262c359da34","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-22 15:29:08.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-850831-9/ns5_nfs : FAILURE in 9h 7m 53s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-850831-9/ns5_iscsi : FAILURE in 1h 25m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"397ef296953558f72b72fc3479f3cb988d8b52bf","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-02-22 15:32:09.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/850831/10/20230222194612/NEC-V-FC : SUCCESS in 2h 57m 16s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/850831/10/20230222194612/NEC-V-ISCSI : SUCCESS in 4h 45m 53s","accounts_in_message":[],"_revision_number":10},{"id":"8d68e43dc639ea8a5c1466ae5bb1d4161bb32a47","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-02-22 15:58:09.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/850831-10/HPMSA-iSCSI/ : SUCCESS in 1:42:12\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/850831-10/HPMSA-FC/ : SUCCESS in 1:17:58","accounts_in_message":[],"_revision_number":10},{"id":"37ad87b803dd02321ec5d16fd0f5401019de6b7b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-22 16:07:06.000000000","message":"Patch Set 10:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-850831-10/ns5_nfs : FAILURE in 56m 48s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-850831-10/ns5_iscsi : FAILURE in 9h 15m 10s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"9af63a316c0f140f51ae21e43848c049ac28ab56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-02-22 16:09:20.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"bc71d70b67fb2d18f2641fa11781ef68f8b4d632","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 16:21:17.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db3c955ef2f54247a32d74db4650974e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/610de4c658274be0bd54764e877e610f : SUCCESS in 9m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5fcb706a3f824105adcd3cb8d657fefe : SUCCESS in 9m 40s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"fb082452bf8660809b16300eb40a5bca6330ad55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-02-22 16:23:11.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"2e7e32d9544db7255169ad33830373f9ebcf2061","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-02-22 16:24:23.000000000","message":"Patch Set 10:\n\nBuild failed.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/850831-10/Seagate-iSCSI/ : FAILURE in 1:58:21\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/850831-10/Seagate-FC/ : FAILURE in 1:27:24","accounts_in_message":[],"_revision_number":10},{"id":"94183c753034bef9b6c5d02e2201d2fd8f09e45c","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-02-22 16:25:43.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/850831-10/PVME-iSCSI/ : SUCCESS in 1:58:06\n- PVME-FC https://os-logs.seagate.com/dell-ci/850831-10/PVME-FC/ : SUCCESS in 1:29:04","accounts_in_message":[],"_revision_number":10},{"id":"790bfc0987798fe950f80e6b35343849d2a69787","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-02-22 16:26:35.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"f4c272716b8d47db5c4f3878f13b1cb309940cb8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 16:34:03.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d9ef8f1efa345a48d47d48c057ffca6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b1265f20f6b4373bfd5155cfa310e78 : SUCCESS in 9m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f80828bfe296497f92c77e840d7f57fc : SUCCESS in 9m 30s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"84d5d3179ad2e39d289feba3e2b6dcf045fa0009","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-22 16:49:30.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/850831 : FAILURE in 23m 33s","accounts_in_message":[],"_revision_number":12},{"id":"f28b5b2f69f3e4734c0bed9ffce6a9a2690bbf80","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-22 16:51:25.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/31/850831/12/check/iscsi-driver-alletra-6k/c8125b2/c8125b2.html : FAILURE in 26m 06s","accounts_in_message":[],"_revision_number":12},{"id":"701e0653cf6925b2b9d1428cd5298e4336796dd9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-22 16:51:58.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850831 : FAILURE in 25m 56s","accounts_in_message":[],"_revision_number":12},{"id":"c943bec5abe5a9ecbd634a0c4c86f9d4bf2388e1","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-22 17:32:10.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/31/850831/12/ : FAILURE in 1h 01m 05s","accounts_in_message":[],"_revision_number":12},{"id":"cbdc80507febc727f02230bb7ad4478fd56663b7","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-22 17:40:19.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3d20fce41f1f4063beedf0aa6bc7010f : SUCCESS in 1h 15m 22s","accounts_in_message":[],"_revision_number":12},{"id":"e4f0e1aee3efa4ea6013b6e6e73ad12eeb389bf7","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-02-22 17:50:51.000000000","message":"Patch Set 12:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/31/850831/12/check/fujitsu-eternus-dx-iscsi/4371fbd : SUCCESS in 1h 26m 57s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/31/850831/12/check/fujitsu-eternus-dx-fc/d617c9d : FAILURE in 19m 05s","accounts_in_message":[],"_revision_number":12},{"id":"3128046852d3f91cfa67e6879fdc0036ad1ce9fc","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-02-22 18:04:52.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"a81596eb375dc043b2dd8d9e6e026652e23a5ce2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-22 18:33:41.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/31/850831/12/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/9e5c9c4/ : SUCCESS in 2h 01m 25s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/31/850831/12/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0575af2/ : FAILURE in 2h 08m 20s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/31/850831/12/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/2db7511/ : SUCCESS in 1h 42m 30s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/31/850831/12/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/e884699/ : SUCCESS in 1h 37m 11s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8a5365fd8f1cfa9416296bdaa8a5c2e1ab783211","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-02-22 21:00:05.000000000","message":"Patch Set 12:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/850831/12/fc : SUCCESS in 0h 45m 54s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/31/850831/12/iscsi : SUCCESS in 2h 01m 07s","accounts_in_message":[],"_revision_number":12},{"id":"6259c317aefc55175f72e5b327c8c4a170d499f4","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-22 21:11:55.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/66d67153ca0a47aeaa239cfb6d3fce9d : SUCCESS in 1h 08m 48s","accounts_in_message":[],"_revision_number":12},{"id":"254cfe4790a1837e8f35c28eb1b050d3e0a584fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 22:10:09.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9b24275464241f48c73aa2925d20b3d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c536116ce9b24116b8d7ce8fcf0d2b06 : SUCCESS in 8m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4db7ffb5dcc40d3b37a92a580e3507d : SUCCESS in 11m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c51df32428f401fa2c0759a48f3d0c2 : SUCCESS in 12m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f810713335ae4f888de037edebb5300e : SUCCESS in 11m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/2d858503a34047bab1563620473ace15 : SUCCESS in 47m 11s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/eadbd16388aa4e8f85aa3ebd5b0e840e : SUCCESS in 1h 02m 51s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5b1b41937e4f43eeb8d8865cde28b80d : SUCCESS in 1h 26m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7353035a0cfb487abdaa698d02744398 : SUCCESS in 48m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0a25c9b81f747fbb4fb01c8a5f9987f : SUCCESS in 7m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4068a090ea684c21ac5775a6a3de3ff4 : SUCCESS in 18m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9c1a0e9a1a214ef4adea7adebd10888d : SUCCESS in 9m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/87e9e15527524c2aa00c5372378ad854 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ee1975def434499390f22b042389d104 : SUCCESS in 5m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b59d94e840c64d248f6d292925967d6c : SUCCESS in 5m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/96953b5b7eda493d96dcfab1101afcfc : SUCCESS in 37m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c9c7e15ee49548e5a7709b58d23f8e88 : FAILURE in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/14d6655d67204451bf92db78298a7eba : SUCCESS in 1h 28m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e20c26252b7f44c3a375a31a2c3963bf : FAILURE in 1h 43m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7ca6af735df44e0d8b93ebb50171364c : SUCCESS in 1h 39m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5b78a241b6644e7f9fcce4506bec0cee : SUCCESS in 1h 13m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/08e8e43af5ad43cba3dc16c7b26ab679 : SUCCESS in 1h 33m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4338daf8ffa1489bb2126a33129d1ab1 : SUCCESS in 44m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/54fc2282ca2c46278680a312a986a27b : SUCCESS in 1h 18m 38s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fa3b37a15ef34e15868c51d65c8d6c04 : SUCCESS in 2h 40m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6ee07e7845f40d7a92b28539c82833e : SUCCESS in 52m 51s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c52e3c7099494315a238a48d68ad361d : SUCCESS in 1h 29m 58s","accounts_in_message":[],"_revision_number":12},{"id":"5ca013683e54a7b9831ec0d40afb7302d871eaf2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-02-22 22:18:18.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/22_02_2023_22_10_18\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":12},{"id":"5c82aaad86a49530e8a677c9e0e23ce05d054b87","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-02-22 22:22:40.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/31/850831/12/20230223030931/NEC-V-FC : SUCCESS in 2h 44m 07s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/31/850831/12/20230223030931/NEC-V-ISCSI : SUCCESS in 4h 13m 06s","accounts_in_message":[],"_revision_number":12},{"id":"34e55a0c09a4316b81d0b096a1aef97e9c91a2d6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-22 23:23:11.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/850831/12/upstream-check/cinder-tempest-plugin-ontap-fcp/d268ba2/ : SUCCESS in 1h 12m 07s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/850831/12/upstream-check/cinder-tempest-plugin-ontap-iscsi/bd0ad0a/ : SUCCESS in 1h 12m 19s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/850831/12/upstream-check/cinder-tempest-plugin-ontap-nfs/e283e08/ : SUCCESS in 44m 41s\n- cinder-tempest-plugin-ontap-nvme-tcp finger://managesf.services.openstack.netapp.com:7979/6e6557b282fb48e5afe6da7e39e26f17 : RETRY_LIMIT in 34m 51s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"9807567fca1e522d91c5c30136772a4928c19518","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-02-23 00:22:06.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/31/850831/12/check/DellEMC_PowerFlex/499bb5b/DellEMC_PowerFlex/1645 : SUCCESS in 1h 43m 23s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"d1b1bc49fa4b5af91e79d21fa8db19e0a5db1e42","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-23 03:08:28.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/850831/12/check/DellEMC_Unity_FC/3d37f41/DellEMC_Unity_FC/2350 : FAILURE in 1h 45m 50s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/850831/12/check/DellEMC_Unity_iSCSI/39f4526/DellEMC_Unity_iSCSI/2203 : FAILURE in 2h 02m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"b1af70e37e18b098fc9f747233ef3e4edcd4082b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-02-23 03:45:16.000000000","message":"Patch Set 12: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"fdeb4e6500f2f7b7c9e82cebec277c47904a22ab","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-02-23 03:53:58.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"27fcc33b7fc53c35266962fedf5d36cdd495ca26","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-02-23 03:54:29.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"df56b39bf560773d3e3b4001850d9c71d9f805cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 03:54:37.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"8ae7ece1db2248baf53718a116ea0167cecf93fa","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-23 04:22:28.000000000","message":"Patch Set 11:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-850831-11/ns5_nfs : FAILURE in 12h 12m 36s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-850831-11/ns5_iscsi : FAILURE in 1h 5m 8s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":11},{"id":"fc6a29e10300b056ab60a15aa61f68ee8ba69bd5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-23 05:09:26.000000000","message":"Patch Set 12:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":12},{"id":"e678d9e198d0abe430e9787ec8cb20dd6bdc39b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 05:16:36.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2eeb6459ce0543fa9059633069d168b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96ef796ed6e4406e9257a5e05320e6c8 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b07fbb208108460faa4289ffa9f3dc3a : SUCCESS in 10m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18d15c4ef8d345b0b90b109cf4300813 : SUCCESS in 10m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/816392a43a38483abdc30c407bf7f607 : SUCCESS in 13m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/9e0415d08b3b4700b46388cc08343e17 : SUCCESS in 41m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6a418d9430ad4d6b8e86d1d26cec8e85 : FAILURE in 1h 03m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/350bdb6bbe474ef4b827cf2ce074e6e5 : SUCCESS in 1h 04m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17d008042bcb4cc7adbeafc1211218be : SUCCESS in 8m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c201d4438e764fc08bb000c6999390e5 : SUCCESS in 54m 15s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d0a2dcf640ce4d278bd6da316a3b37e9 : SUCCESS in 1h 05m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5720503149f448bb754fe03356f9eb1 : SUCCESS in 39m 49s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b73b86aea2442bea2646d790049b094 : SUCCESS in 1h 08m 19s","accounts_in_message":[],"_revision_number":12},{"id":"8b046995a158d37993bf634b6c519be4ec11c5dd","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-23 07:53:08.000000000","message":"Patch Set 12:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-31-850831-12/ns5_nfs : FAILURE in 12h 16m 57s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-31-850831-12/ns5_iscsi : FAILURE in 15h 20m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":12},{"id":"bc7f6262ed1b0e28347de19ce5d95c40848053ea","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-02-23 18:39:36.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/850831-12/HPMSA-iSCSI/ : SUCCESS in 1:38:51\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/850831-12/HPMSA-FC/ : SUCCESS in 1:17:38","accounts_in_message":[],"_revision_number":12},{"id":"0b18915b1fdb8961183622f7f800932e3fc1420d","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-02-23 22:57:59.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/850831-12/PVME-iSCSI/ : SUCCESS in 2:00:18\n- PVME-FC https://os-logs.seagate.com/dell-ci/850831-12/PVME-FC/ : SUCCESS in 1:36:11","accounts_in_message":[],"_revision_number":12},{"id":"c734b2b5dd5940b146503a1f1ccd6a119d0a4f6b","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-02-23 22:58:23.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/850831-12/Seagate-iSCSI/ : SUCCESS in 1:59:04\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/850831-12/Seagate-FC/ : SUCCESS in 1:35:17","accounts_in_message":[],"_revision_number":12},{"id":"3cfef057f8589ddd87283640b6c18211ff13c0fc","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-02-24 15:04:08.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"42b0f2c202aa54bbf880510f8d1d12cd6bf4cb53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 15:37:39.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edf1e55ba40c40a18ae2b52a75a14e69\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d08920630f17458f86f76c0ca94f2242 : SUCCESS in 29m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f4505b735aa4405e8d36a72b04251027 : FAILURE in 30m 43s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"29261c64ad58263740fdd61b375ec4643f5275c8","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-24 16:08:14.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/74ac5a6413484aacb99c71b5ed156869 : SUCCESS in 1h 03m 17s","accounts_in_message":[],"_revision_number":12},{"id":"fc78beffb79c06bb313c7fbd4cc0e7f0f6bfbd12","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-24 16:12:38.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2322235ec19f4c3dba6919bf75762c67 : SUCCESS in 1h 06m 38s","accounts_in_message":[],"_revision_number":12},{"id":"ce6c3eba3e45b1b5722689be9e09e113c40c0d8c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-24 16:45:16.000000000","message":"Patch Set 12:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/fef5d808fcf84e0f84815ee177511f46 : SUCCESS in 1h 28m 58s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/ecfa4f38f6714942a331e7c28aa16411 : FAILURE in 1h 14m 39s","accounts_in_message":[],"_revision_number":12},{"id":"ac2e32d3691fbeeaeee19f6632b0c29c2c678144","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 17:45:20.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a80c5fc9ba934a52b1c0c94b244abfe4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66533d024b6346208ad8b0097779cfc5 : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea94b41dd7414611868d6dcc8484a43f : SUCCESS in 8m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c581f946e8314298ba1936b9da93550e : SUCCESS in 12m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4fea546ff6842fa93b6a7ef257f9722 : SUCCESS in 13m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/d7ed7fd467eb43d8b6959df5a81ef511 : SUCCESS in 1h 14m 29s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/905a2f06a1854cb0ae397784d14dbbec : SUCCESS in 47m 46s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4544989da1434e9f9d9afca02c8dc83f : FAILURE in 1h 26m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/477b96ad66a9413ab0c9ebff2305699d : SUCCESS in 57m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4117c496b75c4bfb95c3a780b26e3764 : SUCCESS in 8m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/05b56a85202946e1a484d92433fbe646 : SUCCESS in 14m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/039901fb8baa4faea5239c7b36104a9a : SUCCESS in 5m 55s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/00b738696fe740af946a2dd92db98729 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0da5646331de46a2b30cc1d2e36a4bb9 : SUCCESS in 4m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fa7e6e32a8d64fd8af11c939a9eceaa9 : SUCCESS in 6m 28s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4d69d6ebc09b4c83ac646106bd00dd69 : SUCCESS in 54m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/596c017d1fd746c68e3fcbbfa5b39949 : FAILURE in 6m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7b2f1d5f2d624db4b233099c2f307f13 : TIMED_OUT in 2h 04m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ac2c340f211c407fb34777684e8dfa81 : FAILURE in 1h 35m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a05930df141345bfac82da2764837a9b : SUCCESS in 1h 55m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f37f1d5b57104837bb8c43266487de46 : SUCCESS in 1h 15m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/81127221fb6249379ae6b0b59cffa639 : FAILURE in 1h 32m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1903618009424060b8c77fad35d9ad76 : SUCCESS in 35m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4d1b2cb858984537aa47e6c9a2eb0900 : SUCCESS in 1h 19m 23s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/105d92045836489fa6c1683659642bc5 : SUCCESS in 2h 33m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef306078ef0248e5b3f7ae21476cbef6 : SUCCESS in 1h 02m 13s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b399c1cbcf34838a387cd1bef0e5d1f : SUCCESS in 1h 18m 13s","accounts_in_message":[],"_revision_number":12},{"id":"ad66dab2997e94a1fd744088ce97b129b3934f12","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-02-24 20:48:39.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"66a024cbbb054d8b9b26d52bdf28dcca8ef68927","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 21:33:22.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26d7f9ce62f0457888dc6e19f5dc0d41\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/da123a2259d3458c9496285527db9f83 : FAILURE in 43m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/74588af5e83149d385bbf227aea5158d : SUCCESS in 11m 41s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"c4d50a2ecb284b2750cbdd60b7c21c4cc6b2ad34","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-24 21:56:09.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/59b4a03001614ec0a364598b0a3784de : SUCCESS in 1h 05m 44s","accounts_in_message":[],"_revision_number":12},{"id":"5204a79723e9c8f81752d33391033a6465dde757","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-24 22:04:12.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/b16279ebd59e441597afe8fbb7743ca6 : SUCCESS in 1h 14m 50s","accounts_in_message":[],"_revision_number":12},{"id":"7f2c46ec427923f0d64027a0f06ff82073b7ca1b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-24 22:23:05.000000000","message":"Patch Set 12:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/9154c702cc2346f2bb734daf50f06e2f : SUCCESS in 1h 19m 14s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/79588920253146558abd4d6353057c6c : FAILURE in 1h 15m 26s","accounts_in_message":[],"_revision_number":12},{"id":"9ce77468099cea0a644e150bc0ce6495d50fc700","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 23:28:17.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8ae4425816e4f3fbe7a9124c329541c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/002d355a90ce458cb69b748abcc9435c : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6aac36923de4f25b2e2d615692129f0 : SUCCESS in 12m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/434b0361627444449089f10627351f9b : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e05ee1f33c84c5f8c35387a4c162c0e : SUCCESS in 14m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/9c6a4ba5f3c14f719c38e1f0f3d026cc : SUCCESS in 52m 24s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6e35ee9bd4f94d5d97ab910664aac076 : SUCCESS in 53m 54s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/18d27476d5584247b25ba09625d81d4c : SUCCESS in 1h 14m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a11f6deb6214fbfb9ead47b5e24b960 : SUCCESS in 53m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6cb8548cddb43549172f66bebd4ce75 : SUCCESS in 7m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/519ff546a905436a86c6e865bb6228ba : SUCCESS in 15m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e6880fec646b4bcd815166e53cad6fec : SUCCESS in 5m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8fa4a80a27b84a99b8c0b6acefedfc8c : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/37d3bfccb2c54420bc5d7f1ac11083b6 : SUCCESS in 7m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9fbccaa0c22e406a99ff845c53d3011c : SUCCESS in 7m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/992311af1eb54c0e95721ed8aace9cca : SUCCESS in 42m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bef1dedb576f4c5d8d44916a414ada74 : FAILURE in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/34b71ada801b46ff8d385080fea926ef : SUCCESS in 1h 15m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e32f2dc6e64a40aea053fd12b525941a : FAILURE in 1h 19m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1515d8fe44264d919eaff08c26ce53c7 : SUCCESS in 1h 53m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c6c226cb682643799a0a3e65e64b4ea4 : SUCCESS in 1h 15m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/17ff9e0f612d49efa3b6cfd4175e9a4f : SUCCESS in 1h 27m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8fe3ec3dc4d64495aa5b81bd46864ac0 : SUCCESS in 44m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/11bce64056394384bdf1e2e05c3b47d7 : SUCCESS in 1h 14m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/161bb0e7e6474e11b3cb75815b6e574b : SUCCESS in 2h 20m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f418152627814868aa10c3a67f39de59 : SUCCESS in 56m 52s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee0e4dec32734ad08a59fbce7d30e014 : SUCCESS in 1h 23m 23s","accounts_in_message":[],"_revision_number":12},{"id":"b8dbc6f90cf7db21ed140c321f88d90953790dad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 23:28:59.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"18aaca515dd9879b2de842cedb3c03a163fcc93f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-02-24 23:36:10.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_02_2023_23_28_27\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":12},{"id":"d63c3486129d0b43cf9c8fbd96d5298f5916c4a5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-25 00:52:05.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/850831/12/upstream-check/cinder-tempest-plugin-ontap-fcp/107b5a6/ : SUCCESS in 1h 03m 17s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/850831/12/upstream-check/cinder-tempest-plugin-ontap-iscsi/2c0065c/ : SUCCESS in 1h 23m 06s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/850831/12/upstream-check/cinder-tempest-plugin-ontap-nfs/c959aac/ : SUCCESS in 1h 18m 45s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/850831/12/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/917fad0/ : FAILURE in 1h 17m 39s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"b4249c01f250ae522574a98e020cd254e7383909","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-02-25 00:58:52.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/31/850831/12/check/DellEMC_PowerFlex/548b81b/DellEMC_PowerFlex/1654 : SUCCESS in 1h 30m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"de16ba7b0913b5d1d0ef5a15040c93d65b83a70d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 01:02:38.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03d6c48f0a414a91bcea95013ab100e4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccfe9ce751e2458a9be4c2c62efe7d4b : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f5ef8ddd2b24197a63e4d70da55f96b : SUCCESS in 11m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b7b585407d84b9aa5cfef6a0a3a8f2a : SUCCESS in 10m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eac41a23ee446a28f60c15426244cec : SUCCESS in 13m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/2534e666ca784e788dc703d73eb3dec7 : SUCCESS in 1h 06m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/67443e769ca54d558f713dccf3cda218 : SUCCESS in 1h 21m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6e3af21fa1d4c268eef5a8d3c369b89 : SUCCESS in 1h 03m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41b724b6806a413eacd09ea7acba65e6 : SUCCESS in 7m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d0dd0579e07c4dc3988b07b8c13fe1a7 : SUCCESS in 1h 10m 20s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1eca1b2ed42d47d0a3617224adf3008d : SUCCESS in 1h 16m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e38493d5e874e549b48ec313cfadbfd : SUCCESS in 51m 43s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efdafc0bdd2d424fa6b0850c89b5f137 : SUCCESS in 1h 28m 19s","accounts_in_message":[],"_revision_number":12},{"id":"165a99383264adde9df40c6e282d1e233270f123","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 01:02:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"7508ed7c3202365631f9057c9ba9df7d76fd56c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-25 01:03:52.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16ecad1b4a9d4103ba148568325eb459\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6251fbce9e8144478c9c431cdba331e0 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70f0c1c62a7e4e028e08ecccb1e74a03 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":12},{"id":"471a45d1eaec8f4817de6f4279591404d4518568","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-25 01:22:32.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/31/850831/12/check/DellEMC_Unity_FC/2c6c4b5/DellEMC_Unity_FC/2366 : FAILURE in 1h 41m 58s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/31/850831/12/check/DellEMC_Unity_iSCSI/68a0e98/DellEMC_Unity_iSCSI/2218 : FAILURE in 1h 53m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"9fdd3f9996ee540a544890b05d3bdf1941c6e267","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-02-25 04:52:34.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/850831-12/HPMSA-iSCSI/ : SUCCESS in 1:38:51\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/850831-12/HPMSA-FC/ : SUCCESS in 1:18:36","accounts_in_message":[],"_revision_number":12},{"id":"04a2f7f8fc529f4fe3ddd0560ab3b9059ca428d0","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-02-25 06:01:12.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/850831-12/Seagate-iSCSI/ : SUCCESS in 1:58:41\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/850831-12/Seagate-FC/ : FAILURE in 1:33:52","accounts_in_message":[],"_revision_number":12},{"id":"ad707daa903e6d4c56280e756d5a3640c0f12660","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-02-25 06:04:56.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/850831-12/PVME-iSCSI/ : SUCCESS in 2:01:57\n- PVME-FC https://os-logs.seagate.com/dell-ci/850831-12/PVME-FC/ : SUCCESS in 1:34:46","accounts_in_message":[],"_revision_number":12},{"id":"3b104ec734c32dd8c3fe6f575d42f61552b5f541","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-01 08:05:21.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"5b38bd138c6f43abdd1f16068cd7ec92f1991007","revisions":{"57bc340425f148dd3f87668bfcb177c4467b44e6":{"kind":"REWORK","_number":1,"created":"2022-07-25 04:48:01.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/31/850831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/850831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/850831/1"}}},"commit":{"parents":[{"commit":"f8fa57a8b974f30cdda8367ebd470279871ea136","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8fa57a8b974f30cdda8367ebd470279871ea136"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-07-23 01:30:41.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-07-23 01:30:41.000000000","tz":0},"subject":"Hitachi: support data deduplication and compression","message":"Hitachi: support data deduplication and compression\n\nSupport data deduplication and compression, by storage assist.\nThe feature can be worked, if user enable deduplication and compression\nfor the DP-pool, by Configuration Manager REST API,\nand set the extra spec \"hbsd:capacity_saving\" to \"deduplication_compression\"\n\nImplements: blueprint hitachi-vsp-support-dedup-comp\nChange-Id: Ia6dcf8ee2205d32714849c0730c9a3562fa8f485\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/57bc340425f148dd3f87668bfcb177c4467b44e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/57bc340425f148dd3f87668bfcb177c4467b44e6"}]},"branch":"refs/heads/master"},"6f8d8c925de98d1407bf72b0581b62e756d3904c":{"kind":"REWORK","_number":2,"created":"2022-09-02 04:43:04.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/31/850831/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/850831/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/850831/2"}}},"commit":{"parents":[{"commit":"469fa370a2f78726ae12873ef79fd220ae37e0ca","subject":"Merge \"Update nova microversion for volume backed instance rebuild\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/469fa370a2f78726ae12873ef79fd220ae37e0ca"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-09-02 03:09:23.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-09-02 03:09:23.000000000","tz":0},"subject":"Hitachi: support data deduplication and compression","message":"Hitachi: support data deduplication and compression\n\nSupport data deduplication and compression, by storage assist.\nThe feature can be worked, if user enable deduplication and compression\nfor the DP-pool, by Configuration Manager REST API,\nand set the extra spec \"hbsd:capacity_saving\" to \"deduplication_compression\"\n\nImplements: blueprint hitachi-vsp-support-dedup-comp\nChange-Id: Ia6dcf8ee2205d32714849c0730c9a3562fa8f485\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6f8d8c925de98d1407bf72b0581b62e756d3904c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6f8d8c925de98d1407bf72b0581b62e756d3904c"}]},"branch":"refs/heads/master"},"b4b17f66fc0b5a6851c561777d37736911bad705":{"kind":"REWORK","_number":3,"created":"2022-09-02 04:45:16.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/31/850831/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/850831/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/850831/3"}}},"commit":{"parents":[{"commit":"469fa370a2f78726ae12873ef79fd220ae37e0ca","subject":"Merge \"Update nova microversion for volume backed instance rebuild\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/469fa370a2f78726ae12873ef79fd220ae37e0ca"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-09-02 03:09:23.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-09-02 04:45:15.000000000","tz":0},"subject":"Hitachi: support data deduplication and compression","message":"Hitachi: support data deduplication and compression\n\nSupport data deduplication and compression, by storage assist.\nThe feature can be worked, if user enable deduplication and compression\nfor the DP-pool, by Configuration Manager REST API,\nand set the extra spec \"hbsd:capacity_saving\" to \"deduplication_compression\"\n\nImplements: blueprint hitachi-vsp-support-dedup-comp\nChange-Id: Ia6dcf8ee2205d32714849c0730c9a3562fa8f485\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b4b17f66fc0b5a6851c561777d37736911bad705"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b4b17f66fc0b5a6851c561777d37736911bad705"}]},"branch":"refs/heads/master"},"b9cc20b35193bece837d6ee759008292437ced86":{"kind":"REWORK","_number":4,"created":"2022-09-05 09:57:45.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/31/850831/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/850831/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/850831/4"}}},"commit":{"parents":[{"commit":"f0dded4d8ed0e47ecd75ffe800824e709a80f0d4","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f0dded4d8ed0e47ecd75ffe800824e709a80f0d4"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-09-05 07:18:53.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2022-09-05 07:18:53.000000000","tz":0},"subject":"Hitachi: support data deduplication and compression","message":"Hitachi: support data deduplication and compression\n\nSupport data deduplication and compression, by storage assist.\nThe feature can be worked, if user enable deduplication and\ncompression\nfor the DP-pool, by Configuration Manager REST API,\nand set the extra spec \"hbsd:capacity_saving\" to\n\"deduplication_compression\"\n\nImplements: blueprint hitachi-vsp-support-dedup-comp\nChange-Id: Ia6dcf8ee2205d32714849c0730c9a3562fa8f485\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b9cc20b35193bece837d6ee759008292437ced86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b9cc20b35193bece837d6ee759008292437ced86"}]},"branch":"refs/heads/master"},"65ca4408abe92bbf62bf59cbab2b7770802f26da":{"kind":"REWORK","_number":5,"created":"2023-02-09 07:55:53.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/31/850831/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/850831/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/850831/5"}}},"commit":{"parents":[{"commit":"3cdf861befa05a020a4e128c553eda4d32aedfa6","subject":"Merge \"Handle external events in extend volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3cdf861befa05a020a4e128c553eda4d32aedfa6"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-02-09 07:41:11.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-02-09 07:41:11.000000000","tz":0},"subject":"Hitachi: support data deduplication and compression","message":"Hitachi: support data deduplication and compression\n\nSupport data deduplication and compression, by storage assist.\nThe feature can be worked, if user enable deduplication and compression\nfor the DP-pool, by Configuration Manager REST API,\nand set the extra spec ``hbsd:capacity_saving`` to\n``deduplication_compression``\n\nImplements: blueprint hitachi-vsp-support-dedup-comp\nChange-Id: Ia6dcf8ee2205d32714849c0730c9a3562fa8f485\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/65ca4408abe92bbf62bf59cbab2b7770802f26da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/65ca4408abe92bbf62bf59cbab2b7770802f26da"}]},"branch":"refs/heads/master"},"79c57fd09d84b9355fb6950bbedaa386ae9e95ad":{"kind":"REWORK","_number":6,"created":"2023-02-20 15:22:05.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/31/850831/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/850831/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/850831/6"}}},"commit":{"parents":[{"commit":"a92aa06e463f6322e8abdc2f570a801ea82dd1cb","subject":"Hitachi: add GAD volume support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a92aa06e463f6322e8abdc2f570a801ea82dd1cb"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-02-20 15:06:54.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-02-20 15:06:54.000000000","tz":0},"subject":"Hitachi: support data deduplication and compression","message":"Hitachi: support data deduplication and compression\n\nSupport data deduplication and compression, by storage assist.\nThe feature can be worked, if user enable deduplication and compression\nfor the DP-pool, by Configuration Manager REST API,\nand set the extra spec ``hbsd:capacity_saving`` to\n``deduplication_compression``\n\nImplements: blueprint hitachi-vsp-support-dedup-comp\nChange-Id: Ia6dcf8ee2205d32714849c0730c9a3562fa8f485\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79c57fd09d84b9355fb6950bbedaa386ae9e95ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/79c57fd09d84b9355fb6950bbedaa386ae9e95ad"}]},"branch":"refs/heads/master"},"6ebc0e75411526978de3798c69a8dd82592b6061":{"kind":"REWORK","_number":7,"created":"2023-02-21 01:46:32.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/31/850831/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/850831/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/850831/7"}}},"commit":{"parents":[{"commit":"a92aa06e463f6322e8abdc2f570a801ea82dd1cb","subject":"Hitachi: add GAD volume support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a92aa06e463f6322e8abdc2f570a801ea82dd1cb"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-02-21 01:20:47.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-02-21 01:20:47.000000000","tz":0},"subject":"Hitachi: support data deduplication and compression","message":"Hitachi: support data deduplication and compression\n\nSupport data deduplication and compression, by storage assist.\nThe feature can be worked, if user enable deduplication and compression\nfor the DP-pool, by Configuration Manager REST API,\nand set the extra spec ``hbsd:capacity_saving`` to\n``deduplication_compression``\n\nImplements: blueprint hitachi-vsp-support-dedup-comp\nChange-Id: Ia6dcf8ee2205d32714849c0730c9a3562fa8f485\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6ebc0e75411526978de3798c69a8dd82592b6061"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6ebc0e75411526978de3798c69a8dd82592b6061"}]},"branch":"refs/heads/master"},"4cc1cd8c3b8c1fda8aa7034fa5fc29d5627ad5ad":{"kind":"REWORK","_number":8,"created":"2023-02-21 11:56:05.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/31/850831/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/850831/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/850831/8"}}},"commit":{"parents":[{"commit":"a92aa06e463f6322e8abdc2f570a801ea82dd1cb","subject":"Hitachi: add GAD volume support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a92aa06e463f6322e8abdc2f570a801ea82dd1cb"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-02-21 01:20:47.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-02-21 11:50:50.000000000","tz":0},"subject":"Hitachi: support data deduplication and compression","message":"Hitachi: support data deduplication and compression\n\nSupport data deduplication and compression, by storage assist.\nThe feature can be worked, if user enable deduplication and compression\nfor the DP-pool, by Configuration Manager REST API,\nand set the extra spec ``hbsd:capacity_saving`` to\n``deduplication_compression``\n\nImplements: blueprint hitachi-vsp-support-dedup-comp\nChange-Id: Ia6dcf8ee2205d32714849c0730c9a3562fa8f485\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4cc1cd8c3b8c1fda8aa7034fa5fc29d5627ad5ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4cc1cd8c3b8c1fda8aa7034fa5fc29d5627ad5ad"}]},"branch":"refs/heads/master"},"2338d0f24c7d19c3eccafde09424cf5edb1d2e39":{"kind":"REWORK","_number":9,"created":"2023-02-22 06:20:42.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/31/850831/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/850831/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/850831/9"}}},"commit":{"parents":[{"commit":"a92aa06e463f6322e8abdc2f570a801ea82dd1cb","subject":"Hitachi: add GAD volume support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a92aa06e463f6322e8abdc2f570a801ea82dd1cb"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-02-22 06:05:51.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-02-22 06:05:51.000000000","tz":0},"subject":"Hitachi: support data deduplication and compression","message":"Hitachi: support data deduplication and compression\n\nSupport data deduplication and compression, by storage assist.\nThe feature can be worked, if user enable deduplication and compression\nfor the DP-pool, by Configuration Manager REST API,\nand set the extra spec ``hbsd:capacity_saving`` to\n``deduplication_compression``\n\nImplements: blueprint hitachi-vsp-support-dedup-comp\nChange-Id: Ia6dcf8ee2205d32714849c0730c9a3562fa8f485\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2338d0f24c7d19c3eccafde09424cf5edb1d2e39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2338d0f24c7d19c3eccafde09424cf5edb1d2e39"}]},"branch":"refs/heads/master"},"b9fa392ab3d1afe8023fd45dbbc3ed47e753f7dc":{"kind":"REWORK","_number":10,"created":"2023-02-22 06:51:26.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/31/850831/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/850831/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/850831/10"}}},"commit":{"parents":[{"commit":"a92aa06e463f6322e8abdc2f570a801ea82dd1cb","subject":"Hitachi: add GAD volume support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a92aa06e463f6322e8abdc2f570a801ea82dd1cb"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-02-22 06:36:38.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-02-22 06:36:38.000000000","tz":0},"subject":"Hitachi: support data deduplication and compression","message":"Hitachi: support data deduplication and compression\n\nSupport data deduplication and compression, by storage assist.\nThe feature can be worked, if user enable deduplication and compression\nfor the DP-pool, by Configuration Manager REST API,\nand set the extra spec ``hbsd:capacity_saving`` to\n``deduplication_compression``\n\nImplements: blueprint hitachi-vsp-support-dedup-comp\nChange-Id: Ia6dcf8ee2205d32714849c0730c9a3562fa8f485\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b9fa392ab3d1afe8023fd45dbbc3ed47e753f7dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b9fa392ab3d1afe8023fd45dbbc3ed47e753f7dc"}]},"branch":"refs/heads/master"},"5506032114b582af9a070a655048db61043d4bd4":{"kind":"REWORK","_number":11,"created":"2023-02-22 16:09:20.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/31/850831/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/850831/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/850831/11"}}},"commit":{"parents":[{"commit":"a92aa06e463f6322e8abdc2f570a801ea82dd1cb","subject":"Hitachi: add GAD volume support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a92aa06e463f6322e8abdc2f570a801ea82dd1cb"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-02-22 15:53:31.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-02-22 15:53:31.000000000","tz":0},"subject":"Hitachi: support data deduplication and compression","message":"Hitachi: support data deduplication and compression\n\nSupport data deduplication and compression, by storage assist.\nThe feature can be worked, if user enable deduplication and compression\nfor the DP-pool, by Configuration Manager REST API,\nand set the extra spec ``hbsd:capacity_saving`` to\n``deduplication_compression``\n\nImplements: blueprint hitachi-vsp-support-dedup-comp\nChange-Id: Ia6dcf8ee2205d32714849c0730c9a3562fa8f485\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5506032114b582af9a070a655048db61043d4bd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5506032114b582af9a070a655048db61043d4bd4"}]},"branch":"refs/heads/master"},"5b38bd138c6f43abdd1f16068cd7ec92f1991007":{"kind":"REWORK","_number":12,"created":"2023-02-22 16:23:11.000000000","uploader":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"ref":"refs/changes/31/850831/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/31/850831/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/31/850831/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/31/850831/12"}}},"commit":{"parents":[{"commit":"a92aa06e463f6322e8abdc2f570a801ea82dd1cb","subject":"Hitachi: add GAD volume support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a92aa06e463f6322e8abdc2f570a801ea82dd1cb"}]}],"author":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-02-22 16:08:39.000000000","tz":0},"committer":{"name":"Atsushi Kawai","email":"atsushi.kawai.bu@hitachi.com","date":"2023-02-22 16:08:39.000000000","tz":0},"subject":"Hitachi: support data deduplication and compression","message":"Hitachi: support data deduplication and compression\n\nSupport data deduplication and compression, by storage assist.\nThe feature can be worked, if user enable deduplication and compression\nfor the DP-pool, by Configuration Manager REST API,\nand set the extra spec ``hbsd:capacity_saving`` to\n``deduplication_compression``\n\nImplements: blueprint hitachi-vsp-support-dedup-comp\nChange-Id: Ia6dcf8ee2205d32714849c0730c9a3562fa8f485\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5b38bd138c6f43abdd1f16068cd7ec92f1991007"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5b38bd138c6f43abdd1f16068cd7ec92f1991007"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[]}
