)]}'
{"id":"openstack%2Fcinder~850836","triplet_id":"openstack%2Fcinder~master~Ie178e48958a69ef2b6c329e132de2c772d5c1a40","project":"openstack/cinder","branch":"master","topic":"bug/1819903","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2022-09-15 16:48:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie178e48958a69ef2b6c329e132de2c772d5c1a40","subject":"HPE 3PAR: Fix umanaged volumes \u0026 snapshots missing","status":"MERGED","created":"2022-07-25 07:18:10.000000000","updated":"2022-09-19 12:02:56.000000000","submitted":"2022-09-15 16:48:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":48,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"850836-bug/1819903","meta_rev_id":"705f886dfb23ce02a90d2300f163de9032e8b1a2","_number":850836,"virtual_id_number":850836,"owner":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-15 16:48:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-09-14 13:38:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-12 16:18:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-09-14 13:38:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-25 07:22:27.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-07-25 07:31:46.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-07-25 07:42:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 08:21:16.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-07-25 08:25:35.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 08:30:29.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-07-25 08:58:20.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-07-25 09:09:07.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 09:13:41.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 09:17:18.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 10:47:34.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 12:06:08.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-07-25 12:10:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-26 09:02:12.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-07-26 09:36:17.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-07-26 09:37:09.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-07-26 10:16:56.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-26 23:16:54.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-07-28 16:09:13.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-08-02 14:16:58.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-08-03 09:58:49.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-08-03 13:27:03.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-03 14:52:18.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-08-03 15:57:28.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-08-10 13:54:08.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2022-08-10 18:00:33.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2022-08-12 12:13:40.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-08 13:09:34.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-14 13:38:11.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-09-15 05:01:33.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"}],"messages":[{"id":"1e50eda16dfd5f6385845aa4fd1b1d0424908b25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-07-25 07:18:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c844368b61043575526f13e5f08cfa8460096ca","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-25 07:22:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"eabff57010cf3177f0d312bced9442bafa9d1ca8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-25 07:31:46.000000000","message":"Patch Set 1:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/228","accounts_in_message":[],"_revision_number":1},{"id":"e1c719cc5ae4f6beb5763c0a813d8aae85061743","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 07:42:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/989ad09587dc4b8e848b4a2c46cc7b7e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4fd24a681ec41a5b2f626649d354145 : SUCCESS in 22m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1d3cb667a090413688cbc83fee21dbb7 : SUCCESS in 15m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0a48b3acd5d79de9b487c54c295966f99f23f16","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-25 08:21:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/36/850836/1/check/iscsi-driver-alletra-6k/6378535/6378535.html : FAILURE in 1h 02m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8f1a5b363dc1bb7ca2842fdda44a6aa7653bf9e7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-25 08:25:35.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"4a81d9e2bd0de9fea545d3d806ae6a0eb9af3612","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-25 08:30:29.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850836/1/iscsi/ : SUCCESS in 1h 11m 15s\n- cinder-smb http://cloudbase-ci.com/cinder/850836/1/smb/ : SUCCESS in 1h 11m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d704ac150f32edf288a43e00cf5a3be8e757043d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-25 08:58:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/36/850836/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/198cc70/ : SUCCESS in 1h 38m 24s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/36/850836/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d7bdacf/ : SUCCESS in 1h 38m 45s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/36/850836/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/9b95bd8/ : FAILURE in 31m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b385cbe2794aab107a9f993d3e1733ef64142eb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-25 09:09:07.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/e61b07ed2aad499397793e0890a4d234 : POST_FAILURE in 1h 14m 35s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/f6a85dd206ad4f509a5802bd7675e48c : POST_FAILURE in 1h 48m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1d331dd83e073c3e5efe3e5fd741dbac180c4198","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-25 09:13:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/36/850836/1/check/fujitsu-eternus-dx-iscsi/d6b6343 : FAILURE in 1h 21m 31s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/36/850836/1/check/fujitsu-eternus-dx-fc/910cfd9 : FAILURE in 1h 29m 52s","accounts_in_message":[],"_revision_number":1},{"id":"80ff62fcc3a934a34c92aa2bd3a0abbbd5360862","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-25 09:17:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850836 : SUCCESS in 1h 57m 29s","accounts_in_message":[],"_revision_number":1},{"id":"5607f703cbdd64c2248cb9de30e01982e3882f30","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-25 10:47:34.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-36-850836-1/ns5_nfs : SUCCESS in 3h 27m 37s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-36-850836-1/ns5_iscsi : SUCCESS in 3h 28m 54s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"34fb1741c9e7c667306d49dd8fa535595dd3564a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-25 12:06:08.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/36/850836/1/check/DellEMC_Unity_FC/774fcb9/DellEMC_Unity_FC/1242 : FAILURE in 20m 18s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/36/850836/1/check/DellEMC_Unity_iSCSI/9b932ba/DellEMC_Unity_iSCSI/1146 : SUCCESS in 1h 50m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"ab8a4176b1a221f33e83f876c95f49d890e82349","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 12:10:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf373e62b1fa4d7d90c8c0d2c83f77cc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50a1755e90ea49ddaec551bba22559f5 : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ead0db120f3485bb07a6e37dfd11e71 : SUCCESS in 11m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa7df49f7cc4426db749e0a81ebe4af9 : SUCCESS in 12m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c543498761414e0d8a37da8d14aacc2f : SUCCESS in 12m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/278529e5232e484ba2c67a6b9274c874 : SUCCESS in 13m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/a591cf2ffdf34f68bbbaa2fe81f954f6 : FAILURE in 1h 01m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bd922929a6584c02a57f227b6923b67c : FAILURE in 1h 01m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3ac6b510d4e422090f25cb9a25b60a4 : POST_FAILURE in 1h 01m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/193df351956445508eed83567157bdd8 : SUCCESS in 9m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/91a49adc56a0450c99bc583a2866dd6f : SUCCESS in 15m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/af00720030044b47a16bc00e225d5c61 : FAILURE in 5m 20s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4d1a163c1fa84c8083fac6a20ddada5a : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/38607a20aaa943a496415ac21f5d1540 : SUCCESS in 6m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9a785c2a919e4a3aa03bc5daf36767a1 : SUCCESS in 7m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3af825d387d14dc4919511e91a9d60f8 : SUCCESS in 32m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d5afd1816bb0496db19a9107aca753c0 : FAILURE in 6m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e669b7bb45b944828955b17521115e5b : FAILURE in 1h 00m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bedfd4d7f213459e8795728cebb600d2 : FAILURE in 1h 33m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f379d3611c454615b881615924a27192 : SUCCESS in 2h 03m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dedf1f8e36d349cb8d3c14f1a9608255 : SUCCESS in 1h 01m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/42bc5e6097144d26b0476a418b88c1e2 : FAILURE in 1h 54m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3a91d25ed84b4d00ba3e98b4b8b92ebe : SUCCESS in 43m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c5c67520c5cf4646882bf4073eddd536 : FAILURE in 1h 04m 38s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8c16931d9f6542a889126d665fe9d475 : SUCCESS in 2h 04m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2228de9e3ab44866ab7d2e471539a584 : POST_FAILURE in 1h 00m 26s","accounts_in_message":[],"_revision_number":1},{"id":"99e9b499401d08df555b0ee83cd2bb517dd5c77c","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-07-26 06:42:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69c48ba9ee4fe12d43bac0c0902ffd38f9741c20","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-26 06:50:00.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"c053e51a1c49e2889bb7e567a26490c8dbe5f5d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 07:07:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66fc2f2d0c76400c8a0117bfdac82932\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e9ec360dbc9401d89a4fe406f33a0ff : SUCCESS in 21m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b82cd11f13c044369f03b2f81f9ad9c3 : SUCCESS in 23m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e571a7cec4b4b44af7ca643dd8d9819e9e025ab6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-26 08:09:53.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850836/1/iscsi/ : FAILURE in 1h 26m 29s\n- cinder-smb http://cloudbase-ci.com/cinder/850836/1/smb/ : SUCCESS in 57m 53s","accounts_in_message":[],"_revision_number":1},{"id":"29dc24ee8bb0ef907f065262bfa08e14ab3eb324","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-26 08:33:06.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/d4fa53ecbb874bc0b4eff66ea1e1585e : POST_FAILURE in 1h 31m 41s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/faeaa1dfd1cd48c8b52170ac3538098f : POST_FAILURE in 1h 47m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d4a1841685baad00015d781e85fa32c9714c6568","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 08:54:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ac608272f584900a3b8414929c19b2b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b7b139bf0344a2aab8165a5e3db9e80 : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d26d1116f28f4cacbe86d2c007902b92 : SUCCESS in 12m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/109a6eb369ef43f5913f9f7fba1c6d16 : SUCCESS in 13m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05e4c31973b14d2185c6c6af02fc1c50 : SUCCESS in 13m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eb7995922d1478f9e6786df66a3b562 : SUCCESS in 12m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/e4e411ec736949f4af06f26966930034 : SUCCESS in 58m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/42166153646e4a3eba0e5284c7c95027 : SUCCESS in 1h 38m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/069de782f9e94400aa4f9426d00268e9 : SUCCESS in 51m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb2604917fe4405c907e40be6aba37a8 : SUCCESS in 9m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/444b0e12bdf34be8bd162c5ce2eb4c81 : SUCCESS in 15m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c86e95b8957b49048f2a47462d9f6af3 : FAILURE in 6m 20s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/684a1a73639548b8ad98f343e90b7700 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4f194b969d16438385f70a592bc384c7 : SUCCESS in 7m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8ae8a83095374e2b93ff2bfaf3ce32aa : SUCCESS in 8m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b323e7da315840588f870005f7425117 : SUCCESS in 42m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/77d95e6ed31c44638adbb495c112e418 : FAILURE in 7m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e29c5eecb97041b89801bf5eb83668f8 : SUCCESS in 1h 15m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e7c6ccf9dc144fe18977eac14e1e0063 : FAILURE in 1h 49m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1ab8947eaf124d6aba7b278821b8c58b : SUCCESS in 1h 34m 29s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bc91ba1e52224d7e8eb74ee8749648fe : SUCCESS in 1h 09m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/095afebeaab34fd2947d4aab75bcb1c2 : FAILURE in 1h 47m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4b40a0fd06fb4a5f87d8cbe77cddc31b : SUCCESS in 41m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5d331a64341a4d97bc6c95cd90d48c2c : SUCCESS in 1h 22m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b2360e82741e4b96a1debfbe3cfa0643 : SUCCESS in 2h 09m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d9034215ebe462b9d0c8441bf96d8a1 : SUCCESS in 52m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7539210af018a12dbe52348c649b7d48e8ca796a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-07-26 09:02:12.000000000","message":"Patch Set 1:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/26_07_2022_08_54_12\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"89cae503a80499e6fcc91305424f60e791283c06","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-26 09:36:17.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/36/850836/1/check/DellEMC_PowerStore_NFS/7e40236/DellEMC_PowerStore_NFS/175 : FAILURE in 41m 51s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"9935402bb559afeddca533b1462e4ac4111355f2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-26 09:37:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/850836/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/c528d5a/ : SUCCESS in 41m 57s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/850836/1/upstream-check/cinder-tempest-plugin-ontap-nfs/e4fa585/ : RETRY_LIMIT in 7m 26s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"22d689975d823bdd3f0377eda5a80f24d535c60a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-26 10:16:56.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/36/850836/1/check/DellEMC_PowerFlex/6030d23/DellEMC_PowerFlex/1073 : SUCCESS in 1h 22m 25s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"160c2c07879e93c7b4127444ac7a2f758c288911","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-07-26 23:16:54.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/36/850836/1/check/DellEMC_SC_FC/3fb68ca/DellEMC_SC_FC/805 : SUCCESS in 51m 44s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/36/850836/1/check/DellEMC_SC_iSCSI/73d1a3e/DellEMC_SC_iSCSI/827 : SUCCESS in 1h 57m 23s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"3e33d30dbbe1cff4cc32212960c7f7a5b6c6f7ef","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-27 06:37:59.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/36/850836/1/check/DellEMC_Unity_FC/99b4822/DellEMC_Unity_FC/1259 : FAILURE in 18m 26s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/36/850836/1/check/DellEMC_Unity_iSCSI/1808b4d/DellEMC_Unity_iSCSI/1160 : SUCCESS in 2h 04m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"ebee301c2446b49a2b6d55af058385b5516c4ab8","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-28 16:09:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/36/850836/1/20220728224643/NEC-V-FC : SUCCESS in 2h 16m 54s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/36/850836/1/20220728224643/NEC-V-ISCSI : SUCCESS in 2h 22m 27s","accounts_in_message":[],"_revision_number":1},{"id":"6538cf16383af4ee7072c6588d9d85235bab2549","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-08-02 12:27:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a45177f230a6ff4f1844a8b28aab933399d4558","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-08-02 14:16:58.000000000","message":"Patch Set 1: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"94bc78e16a84b2ee9925467bed95925416139b68","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-08-02 14:19:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de71597dc71cf9db3c6c36263eb5331fa50110ab","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-08-03 04:10:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e096ce28f0f6cf585d3f7f9d24ff6ace2d1b758","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-08-03 09:58:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"873982dffe2033c1072ba516e5fcefe6960ee52b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-03 09:58:49.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1659520726829 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"5c49cfad04b59a9ef50ec4bee8325c1e0c748331","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-03 10:05:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k finger://openstackcifc-virtual-machine/4577b6d1b7c344a7b79bc940bcf37856 : RETRY_LIMIT in 10s","accounts_in_message":[],"_revision_number":2},{"id":"2a089f6275b40ec7dd685a81e66515fb65fe3a55","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-03 10:16:51.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"9639d0b9ad4c81e7d9ab981f7a2321b06cd0d2b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 10:20:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2592afcbacc40409c51e12b73df5591\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/18956fd4bb184fd79e7d9198bc44a543 : SUCCESS in 15m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/07f616e0b60145cab4041290a929babb : SUCCESS in 16m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2055860ac4d831e7c86ca196b73ed5f5fd63028f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-03 11:00:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850836/2/iscsi/ : SUCCESS in 1h 00m 58s\n- cinder-smb http://cloudbase-ci.com/cinder/850836/2/smb/ : SUCCESS in 1h 01m 25s","accounts_in_message":[],"_revision_number":2},{"id":"dd144c089c6b39d23b1faab4497ca5a93c8e983a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-03 11:36:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/36/850836/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b741c9d/ : SUCCESS in 1h 32m 14s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/36/850836/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d108421/ : SUCCESS in 1h 36m 38s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/36/850836/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/56626eb/ : SUCCESS in 1h 30m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"569fb03bf4023a4b654b76500121116a7495277f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-03 11:47:04.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/36/850836/2/check/iscsi-driver-nimble/9083b2a/9083b2a.html : FAILURE in 1h 46m 57s","accounts_in_message":[],"_revision_number":2},{"id":"f25274054369d3d2947d4f163879d488e06c01e2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-03 11:57:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850836 : SUCCESS in 1h 57m 11s","accounts_in_message":[],"_revision_number":2},{"id":"8040cea5e2f4c9ade0af93f12b8328f814c2a9fe","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-08-03 11:59:19.000000000","message":"Patch Set 2:\n\n(8 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1d03980adde50f3fb086ecf830ea22407c12fd6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-08-03 12:05:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ea895bbb404a9a958d989d0c0b741f7d208369a2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-03 12:05:09.000000000","message":"Patch Set 2:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2},{"id":"bcbaed68ba36a89fa9fe10c6c7a16f7eecfc1188","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-03 12:05:18.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1659528317030 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"5e1ec3610482d40083bec0c1cf3428b74b7d87fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-08-03 12:07:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9c4538ea18554c6cc14b16a9092c27ebd9917117","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-03 12:07:06.000000000","message":"Patch Set 3:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":3},{"id":"01926aa94ad09ae2f721296db4438896fbf2cb8e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-03 12:07:13.000000000","message":"Patch Set 4:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1659528432033 : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"b3fcbd1f93fe6d480a915ad192fc7528cf60b25f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 12:29:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01e659ac1aa846c49b7800a892fb2ec5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5dc939eb67144c199760d959a38f490f : SUCCESS in 15m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/be40e715ff20443d8b78769cba807055 : SUCCESS in 15m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"83b30950a6c27e04c40f2eea77204a670c45c3d5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-03 12:43:58.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"e05ea29d4b4bd246af3483faf791d276610659f4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-03 13:21:44.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850836/4/iscsi/ : SUCCESS in 1h 14m 06s\n- cinder-smb http://cloudbase-ci.com/cinder/850836/4/smb/ : SUCCESS in 56m 53s","accounts_in_message":[],"_revision_number":4},{"id":"6c523483bdbd45d150938bdeac398321770ae8d2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-08-03 13:27:03.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/12fcb2b28d7d4ab3abf98e7d5ed515cf : POST_FAILURE in 34m 42s","accounts_in_message":[],"_revision_number":4},{"id":"e1dec887b8511d76cb4bb4d614e7b2f45338857a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-03 13:27:51.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-36-850836-2/ns5_nfs : SUCCESS in 3h 28m 41s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-36-850836-2/ns5_iscsi : SUCCESS in 3h 27m 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"5ba8993c10f998d8622299e33ce796d328f36544","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-03 13:45:25.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/36/850836/4/check/iscsi-driver-alletra-6k/7a85c17/7a85c17.html : FAILURE in 1h 35m 23s","accounts_in_message":[],"_revision_number":4},{"id":"07cfbab5b7611c46f4862663dfa68a89361e064c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-03 13:50:20.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850836 : FAILURE in 1h 38m 25s","accounts_in_message":[],"_revision_number":4},{"id":"3c8fdb653f394efa20eb02fee553650274762bd6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-03 13:53:20.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/36/850836/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b8f91f6/ : SUCCESS in 1h 30m 24s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/36/850836/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/1d35f40/ : SUCCESS in 1h 44m 21s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/36/850836/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/aec4d4d/ : SUCCESS in 1h 32m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"29fcf308137a0207619ad3e3700173d76f1e5a8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 14:38:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f95753ea9464956a1ad8c46be64b9f6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61ac349d430744fe94dc40f44b5cdc76 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd53a32c3b70479b8a661ec4f80c7cf5 : SUCCESS in 10m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f61c943a57b4497ac9d73eb79c77a80 : SUCCESS in 11m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca977ec6134a4d95bba98970bee77dde : SUCCESS in 12m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0381723fa36a450aa76cabd930d17d1f : SUCCESS in 10m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/2a8776ec44c54564a3d965d39f5a251b : SUCCESS in 45m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f510cbffdc1a4c56b2e9c23cbbfea49b : SUCCESS in 1h 13m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0efcc71f92d14068a8093363287564e4 : SUCCESS in 46m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/544acb2888704b769934bc389ab0a30c : SUCCESS in 7m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2dcab315d9424c52ba6256fd2d1a903c : SUCCESS in 16m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5ffe1368295a49efb5d99b93ea0ff3cb : SUCCESS in 6m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/59e08aed25a043bbbab7d49041c72d55 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ddf0080df9f24d35ba44b7ac8e47d1bf : SUCCESS in 7m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8f00c7e11cbd4090a9436c255ce38913 : SUCCESS in 7m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/af835ea2ec2e47cea5d8d81a7b394e25 : SUCCESS in 37m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/14db69701e0b421aa144bae122e480d1 : FAILURE in 8m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c321d07ebde54eed909160ba81a92cf0 : SUCCESS in 1h 30m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bd51b3f698454381b804ca8a06e8f65d : FAILURE in 1h 43m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/79ddad90ce654be996932392c4ade1b2 : SUCCESS in 1h 32m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b5440f08059e4f67ab7bf579f457bbf6 : SUCCESS in 1h 21m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9168cba2e64d445ebc1eb46313ff45f6 : SUCCESS in 1h 16m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0d2936273a264325989c66a27d0b695d : SUCCESS in 50m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/32fe083f44364d5598bcc2cd1de9fb95 : SUCCESS in 1h 36m 45s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/73bdd8f755464d3a8ce0c9705f18ab68 : SUCCESS in 2h 24m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3eeb284572034cfab68814d6b6de1c4b : SUCCESS in 1h 01m 39s","accounts_in_message":[],"_revision_number":4},{"id":"afa9d218007d64bbddf82dd3c7b7be2a6bf56a32","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-08-03 14:52:18.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4},{"id":"e305f2490ee944da12425a09bdf0e5e705029eb7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-03 15:12:15.000000000","message":"Patch Set 4:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/03_08_2022_14_50_34\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"1bfacde742097febf619cc721a017cf6c262d1a3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-03 15:22:31.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/850836/4/upstream-check/cinder-tempest-plugin-ontap-fcp/7bda6e0/ : RETRY_LIMIT in 8m 19s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/850836/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/68c1583/ : RETRY_LIMIT in 8m 01s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/850836/4/upstream-check/cinder-tempest-plugin-ontap-nfs/0101946/ : SUCCESS in 43m 21s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/850836/4/upstream-check/cinder-tempest-plugin-solidfire-iscsi/24da856/ : RETRY_LIMIT in 8m 04s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"78a0c94344e2aed355585a7ac030a01e41c6ca2b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-03 15:26:19.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/8a22bc7f846a46fdb972ee70d5d33752 : TIMED_OUT in 3h 14m 06s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/e1f435a84ff74560a15cf535efbfeda7 : POST_FAILURE in 3h 08m 33s","accounts_in_message":[],"_revision_number":4},{"id":"4d48a5b3e929f3b0165f11bfac6d7b949ec3bfdd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-03 15:33:23.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/36/850836/4/check/iscsi-driver-nimble/af5886b/af5886b.html : FAILURE in 1h 50m 40s","accounts_in_message":[],"_revision_number":4},{"id":"efc88ff05060ee6614e5a3c65cf5984bbf0c9460","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-03 15:38:11.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-36-850836-3/ns5_nfs : SUCCESS in 3h 29m 27s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-36-850836-3/ns5_iscsi : SUCCESS in 3h 32m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"648cb157a61ad4e19802bf7afcd97cce676b587f","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-08-03 15:57:28.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4},{"id":"ec12666777d1d4db90bed005e93e1fe1e321c9db","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-03 15:59:31.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-36-850836-4/ns5_nfs : SUCCESS in 3h 24m 21s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-36-850836-4/ns5_iscsi : SUCCESS in 3h 31m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"6f58b206aac396347522dc65d516ebe7cafdb726","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-03 16:51:57.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/36/850836/4/check/DellEMC_PowerFlex/249aee1/DellEMC_PowerFlex/1132 : FAILURE in 1h 04m 57s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"360739a671e0ee2803c7f6cca5240a8e3beb8414","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-03 19:45:43.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/36/850836/4/check/DellEMC_PowerStore_FC/629eed5/DellEMC_PowerStore_FC/671 : FAILURE in 17m 15s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/36/850836/4/check/DellEMC_PowerStore_NFS/b484e77/DellEMC_PowerStore_NFS/215 : SUCCESS in 1h 27m 26s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/36/850836/4/check/DellEMC_PowerStore_NVMeTCP/e2f0afc/DellEMC_PowerStore_NVMeTCP/87 : SUCCESS in 2h 03m 52s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"bf9fe45b029e3ead4d1b9108211bc7b930ee035d","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-08-03 19:57:34.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/36/850836/4/check/DellEMC_SC_FC/3a7f0c7/DellEMC_SC_FC/861 : SUCCESS in 50m 03s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/36/850836/4/check/DellEMC_SC_iSCSI/09c99c6/DellEMC_SC_iSCSI/879 : SUCCESS in 2h 03m 30s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"8349faf87b6a7a90e824d3442ddad242ae45e722","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-04 01:19:22.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/36/850836/4/check/DellEMC_Unity_FC/6ef76fd/DellEMC_Unity_FC/1351 : FAILURE in 20m 21s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/36/850836/4/check/DellEMC_Unity_iSCSI/4ebaa00/DellEMC_Unity_iSCSI/1226 : FAILURE in 2h 12m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"42a5cb5aa1f65391aeb1ce6ae9c1df3e95ab007f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-04 05:36:16.000000000","message":"Patch Set 4:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/367","accounts_in_message":[],"_revision_number":4},{"id":"59d98d5dd67ac53dd10f1a8e75595ef860180cd3","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-08-04 06:38:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6157278554b42feb5ab450d069a8566547abf5e5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-08-04 15:01:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/36/850836/4/20220804213714/NEC-V-FC : SUCCESS in 1h 54m 55s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/36/850836/4/20220804213714/NEC-V-ISCSI : SUCCESS in 2h 24m 08s","accounts_in_message":[],"_revision_number":4},{"id":"93d743611a886e82de6d0e4d28801832d3feecd6","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-08-05 04:40:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2abf8af0b5a785cef4a9538ac02c0d1d80bbd6de","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-05 05:04:59.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850836 : FAILURE in 23m 01s","accounts_in_message":[],"_revision_number":4},{"id":"85d3d151fe9aac9ca1eb9471dc95d5f151eaaea1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-08-08 10:36:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"908cccd9a1db5ac40723b60f6a8b19410f5d719d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-08 10:36:51.000000000","message":"Patch Set 5:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1659955008565 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"851b95de421ce5b36c8c23da1d17d089f0de0bb9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-08 10:57:24.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/36/850836/5/check/dsvm-tempest-cisco-zonemanager-job/8fa1bba/ : FAILURE in 8m 16s","accounts_in_message":[],"_revision_number":5},{"id":"3521084c009a6616568507c5e0682d03d26d06e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 11:00:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c4044629e8047cabca3667fb7e0e51e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/50d37903b40f45ad8c1bdcd368c3dc88 : SUCCESS in 16m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/620fa5d8209a435b8294eaaa16766141 : SUCCESS in 21m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b23eddb60e826a7ef524f237c06990a9d00af1ae","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-08 11:39:06.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/36/850836/5/check/iscsi-driver-alletra-6k/7d3c9d6/7d3c9d6.html : FAILURE in 1h 01m 26s","accounts_in_message":[],"_revision_number":5},{"id":"d9b92e658879abb341273620dcf30adb297c9c91","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-08 12:05:18.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850836 : SUCCESS in 1h 26m 52s","accounts_in_message":[],"_revision_number":5},{"id":"4f881b6fac8dc9add1249ae19e8e56d205d49ccc","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-08 12:17:01.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/36/850836/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a1ecec8/ : SUCCESS in 1h 23m 31s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/36/850836/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a807fc8/ : SUCCESS in 1h 39m 25s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/36/850836/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/369d067/ : SUCCESS in 1h 25m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f48f7f2a4eb9beb1affee240c0b8f3162463f6c0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-08 12:27:21.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/36/850836/5/check/iscsi-driver-nimble/fe2339f/fe2339f.html : FAILURE in 1h 46m 37s","accounts_in_message":[],"_revision_number":5},{"id":"4975ebefcb0355aee19c9ea5fff29aee93277e47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 12:41:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/34512e7eefb34d3aa665ca87a21b1b70\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/564aed81eac049dda2502898c2d2c41b : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e17ca19dc544d26ae7ad28728d6b551 : SUCCESS in 10m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce270e06f7b946729dd140f6dfe1cdc5 : SUCCESS in 11m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cba8d0681a2d4a9c992680d51f1c6420 : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0feedfebe1140aba88d76a602574f25 : SUCCESS in 13m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/2e94f32c9a4b4d3f8ded674f2bfcbf9c : SUCCESS in 51m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5dbd8f98c2244f2482838a686029a9e2 : FAILURE in 1h 24m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca88762045c340638708005d33d0dfac : SUCCESS in 39m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8ebf423cfc84a1eb413a2c8828956cd : SUCCESS in 8m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3f53401af5584ddc8dff4b6cdb847e43 : SUCCESS in 14m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c9a3d3a709934fe9a7f3b5aa9b9a36d8 : SUCCESS in 5m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ed82a8a6ad3b43f39f2c3abb69381d37 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dd76bea7e6464ea4a694d8b0acdf2ec2 : SUCCESS in 6m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f836e1be777e4568873075bde3f0a86b : SUCCESS in 6m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bd8904a380d847519352ce03df26f9aa : SUCCESS in 36m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/33de0e5e4ef245b99650a90115a0be67 : FAILURE in 6m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a194273b4bf6446cb6e50fb89b668d47 : SUCCESS in 1h 22m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/14817bad27924a7e9f348e0891c7a282 : FAILURE in 1h 12m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cf436a1508d543149b0c622bd3326c8c : SUCCESS in 1h 51m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/88f0846f05194f328c671bda52db3fa1 : SUCCESS in 1h 15m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/460b23c4079c4ac3b5b52ad14e00d13d : SUCCESS in 1h 25m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/53b57265c5b649cd8b8cf8bc139452ac : SUCCESS in 41m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/865bb39e13714e41a316c10e7f7ef83f : SUCCESS in 1h 28m 17s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0897c641feb1426298ac5565c0b2d5b1 : SUCCESS in 2h 03m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae2101016c2b4852aadc8a1a1684d4c0 : SUCCESS in 56m 49s","accounts_in_message":[],"_revision_number":5},{"id":"6e20d95d65e27b192fe99f863529330ae09d0bdc","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-08 14:10:48.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-36-850836-5/ns5_nfs : SUCCESS in 3h 30m 39s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-36-850836-5/ns5_iscsi : SUCCESS in 3h 33m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"7559f6f86a6e9df999f463deb1564af488ee8bb0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-08 14:38:47.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/5656e6ade57b46c6a604275f83cb8ba9 : POST_FAILURE in 1h 45m 40s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/4c760cea072a4682b8e9f13a29e1e77d : POST_FAILURE in 2h 46m 11s","accounts_in_message":[],"_revision_number":5},{"id":"dec8cd6f16c603dc73d6ae04da0332ce1303a122","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-08 15:02:37.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850836/5/iscsi/ : SUCCESS in 1h 14m 42s\n- cinder-smb http://cloudbase-ci.com/cinder/850836/5/smb/ : SUCCESS in 1h 17m 07s","accounts_in_message":[],"_revision_number":5},{"id":"4646d8c7bea6b2cea67cb44d470be22b4542f238","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-08 19:02:33.000000000","message":"Patch Set 5:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/395","accounts_in_message":[],"_revision_number":5},{"id":"f74d58bcfae4cbe94b73fc70fcba7c8db802d740","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-08 21:24:52.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/36/850836/5/check/DellEMC_Unity_FC/8464a09/DellEMC_Unity_FC/1402 : FAILURE in 18m 57s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/36/850836/5/check/DellEMC_Unity_iSCSI/f224aef/DellEMC_Unity_iSCSI/1287 : FAILURE in 2h 15m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"1214ee25d32d5849fdc286de42d71e0353c24635","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-08-08 22:59:28.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/36/850836/5/20220809052628/NEC-V-FC : SUCCESS in 2h 02m 48s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/36/850836/5/20220809052628/NEC-V-ISCSI : SUCCESS in 2h 32m 56s","accounts_in_message":[],"_revision_number":5},{"id":"550e05164e8ce45ce35c479031124195cdc2893f","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-08-10 04:09:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"673cbd6ff0ecf3b0456341bc0d1ee7d455db4dbf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 04:38:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c4e8cdafb274db9b575ec87d0d50f42\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2e286f695924399a38110535a0829ef : SUCCESS in 14m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/272c5388251c4c7180de0a872f95f21b : SUCCESS in 26m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9ea46a92db84d8bc7e93932f86529c2f3df25c94","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-10 04:51:00.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/36/850836/5/check/dsvm-tempest-cisco-zonemanager-job/0160fc0/ : FAILURE in 34m 35s","accounts_in_message":[],"_revision_number":5},{"id":"b7aac8096f1493636ed3078f42590013bee5f494","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-10 05:31:01.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/5a752b52319144548a94f95652e8cee1 : POST_FAILURE in 1h 11m 07s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/905c9fcc822a4b3d999adcacf494a30e : POST_FAILURE in 1h 18m 10s","accounts_in_message":[],"_revision_number":5},{"id":"bb3db27198beebbf0e47af650d1c581682abfb6a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-10 05:51:42.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850836/5/iscsi/ : SUCCESS in 1h 40m 25s\n- cinder-smb http://cloudbase-ci.com/cinder/850836/5/smb/ : SUCCESS in 1h 41m 24s","accounts_in_message":[],"_revision_number":5},{"id":"6b5a8a0d331c8e16de7033d9b407df1712880a37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 05:54:37.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a979787433c4652a06255c2d826bec5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c303ba12c454ba59340d081d5dbf2e2 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4804b0b7ace04bb7937e353f538032ae : SUCCESS in 10m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9741779bf0e14a9db1638dc0f07a8642 : SUCCESS in 11m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7bece747fef4921891b92323a3d81b1 : SUCCESS in 11m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf1303a2264f43ec9514e7172b583c00 : SUCCESS in 11m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/213260182866471691b287dce8029492 : SUCCESS in 1h 06m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ebc72c1fe8e4436281262845c919c5a9 : SUCCESS in 1h 18m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6d70eb2d323414292daca43cd6eedcd : SUCCESS in 53m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/efb9e9440cfb4bf8b1794a1d330d61ba : SUCCESS in 8m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/93e09efb8e7e4f58ad028c1f4a7efa76 : SUCCESS in 13m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e3b4bf10d3d041bead8a4875534df97e : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7ce33916400d48008ea51e635aed78c6 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1f219e12d1904ce28b9a92ddea462cbb : SUCCESS in 5m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f73ba5df719048639d6054fe45c816df : SUCCESS in 5m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8c62b6ea008146efb4cce079780c0dc1 : SUCCESS in 39m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/523fa75eac0d45aaba6d6b41e16e80d5 : FAILURE in 5m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dfaa66e08f9f439a96507a8ea0aa871a : FAILURE in 1h 05m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0018ac039c0b48c2aad430e6c316ef2a : FAILURE in 1h 17m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/272835065ad541abb4717e655c871205 : SUCCESS in 1h 41m 00s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cfd9efee6c804f2ea92277ac02479a51 : SUCCESS in 55m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/193f5807cbf54a388d7968ffe1dbe7f0 : SUCCESS in 1h 31m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5599f294a2fa424494a0838af0edf078 : SUCCESS in 32m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8ab05481f91c4dab81cdd1d66f85b445 : SUCCESS in 1h 36m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ad84960d1e8d4657b4fde57a3aa3cda8 : SUCCESS in 1h 42m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/250b50abeee240149ea1efd84d12644e : SUCCESS in 1h 00m 03s","accounts_in_message":[],"_revision_number":5},{"id":"2e7fdc764a6197d49a6d43d6aba650e39d65402b","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-08-10 09:20:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bc82a368ea8ff7423acb0255b97d20f201ac48b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 09:39:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55badf05a08446feba56861c012df838\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7db5e04629d54b7c8300bf64bda2686b : SUCCESS in 13m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/20c979617c37434689ae41152869f2c8 : SUCCESS in 14m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"57078c33b3bc132f9532b67cbfb1680b6bd65e7b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-10 10:02:46.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/36/850836/5/check/dsvm-tempest-cisco-zonemanager-job/8212996/ : FAILURE in 34m 43s","accounts_in_message":[],"_revision_number":5},{"id":"78a16d573ee2f50641e8c71831253a44e280cbf2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-10 10:15:41.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850836/5/iscsi/ : SUCCESS in 54m 03s\n- cinder-smb http://cloudbase-ci.com/cinder/850836/5/smb/ : SUCCESS in 52m 56s","accounts_in_message":[],"_revision_number":5},{"id":"f7efd1990cefcb14197eae90d3badfe479ae798a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-10 11:22:47.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/b905890fe4d04601b9d2c77b960e2277 : POST_FAILURE in 1h 59m 20s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/cc2b69a89a894e13a76df1f8142f4dc1 : POST_FAILURE in 1h 53m 46s","accounts_in_message":[],"_revision_number":5},{"id":"bed17a04491ff145f7a0bad28670885e3547cea1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 11:32:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/327461dd941a4d989407b8bddc33845e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/904da4c405974c7399769b02916172bf : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fd1124ba0e24db28f8c82ea8b36cdff : SUCCESS in 12m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97bff511a0f940e391b22fbd4c5b954f : SUCCESS in 12m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46c43f0c9f5e46568bb540e30b901b2f : SUCCESS in 11m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d246507398244579270f8b491bd9d21 : SUCCESS in 10m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/e02e07a192c94e89ac2a0e64adeaffee : SUCCESS in 59m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d2e6cb1e95cf462dbbf22ecec56a9eda : SUCCESS in 1h 15m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d7932fec42042978674c78fc6506233 : SUCCESS in 51m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/713b78233e4d469a91216fdeaedde3e4 : SUCCESS in 8m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3bd2398ef3bc4de0ad50e8a9e10e3f9e : SUCCESS in 15m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e9a966032c384c9c808401b51b320b7d : SUCCESS in 5m 12s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3a3ff450f078461fa1125de72c0150e8 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bf4d717a47e24e5291cbb20557d1be3e : SUCCESS in 5m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dbf2bd3b3811453e823eb17ae306a9d6 : SUCCESS in 6m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/345128bca65149f0a1138e9c2a4174ef : SUCCESS in 40m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2bed56482dc44d288500deab3cbae742 : FAILURE in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b3fc191b449945d1a63896445526190e : SUCCESS in 1h 21m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/677b27a4855a4676bc13b1441b355aa4 : FAILURE in 1h 15m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7247b7cb623543b4b01900f2d03d76ab : SUCCESS in 1h 31m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/881acfb6ae32435daed919c7dcba153a : SUCCESS in 1h 20m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/58eb4d51c2ee4b1186e5a85d746bc217 : FAILURE in 1h 25m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a9b3e43ead014095a5b00abfd913201f : SUCCESS in 37m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7bd57d91f2744bad82918407f090bda5 : FAILURE in 1h 25m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4a5a87b0129842c48af6cfd400e029a4 : SUCCESS in 2h 08m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/425b7535199744599406d3b093e6aa70 : SUCCESS in 41m 49s","accounts_in_message":[],"_revision_number":5},{"id":"6c04d04652f21bb985f256dc6bd0c89b9ffa61ff","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-10 13:03:53.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/850836/5/upstream-check/cinder-tempest-plugin-ontap-fcp/3e785c5/ : RETRY_LIMIT in 10m 31s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/850836/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/0a4b0a5/ : SUCCESS in 1h 10m 28s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/850836/5/upstream-check/cinder-tempest-plugin-ontap-nfs/d790b90/ : SUCCESS in 42m 50s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/850836/5/upstream-check/cinder-tempest-plugin-solidfire-iscsi/f114ab6/ : SUCCESS in 1h 29m 31s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"94a0876345668e36c24fc070a280c701067cb84b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-10 13:29:00.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/36/850836/5/check/DellEMC_PowerStore_FC/fa71567/DellEMC_PowerStore_FC/714 : SUCCESS in 1h 17m 53s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/36/850836/5/check/DellEMC_PowerStore_NFS/8ab38c2/DellEMC_PowerStore_NFS/255 : SUCCESS in 1h 17m 45s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/36/850836/5/check/DellEMC_PowerStore_NVMeTCP/27aba1b/DellEMC_PowerStore_NVMeTCP/129 : SUCCESS in 1h 40m 54s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"6a88b837ab3980eaca93612a2ed7003b67befefe","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-08-10 13:45:47.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/36/850836/5/check/DellEMC_SC_FC/ce9e606/DellEMC_SC_FC/901 : SUCCESS in 53m 00s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/36/850836/5/check/DellEMC_SC_iSCSI/fa825c1/DellEMC_SC_iSCSI/919 : SUCCESS in 2h 12m 44s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"468469bafd9013f3e10d5b787ff61d345616c7ae","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-10 13:49:57.000000000","message":"Patch Set 5:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/10_08_2022_11_32_50\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"c4a0aa13bbb49a5f8eeee41b9abbf5b2380a61f0","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-08-10 13:54:08.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/850836-5/HPMSA-iSCSI/ : SUCCESS in 57:34.40\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/850836-5/HPMSA-FC/ : SUCCESS in 53:19.94","accounts_in_message":[],"_revision_number":5},{"id":"dd3875ba0149fd3179808f4097bec8d47237b34d","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-08-10 14:22:42.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/850836-5/PVME-iSCSI/ : SUCCESS in 1:10:22\n- PVME-FC https://os-logs.seagate.com/dell-ci/850836-5/PVME-FC/ : SUCCESS in 1:08:44","accounts_in_message":[],"_revision_number":5},{"id":"956b4d737f7dfb45a42d2a65d324ea969494432e","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-08-10 17:21:15.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/850836-5/Seagate-iSCSI/ : SUCCESS in 1:18:34\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/850836-5/Seagate-FC/ : SUCCESS in 1:16:06","accounts_in_message":[],"_revision_number":5},{"id":"793174a60c8f3fc3835061823605fee6889bba1b","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-08-10 18:00:33.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/850836-5/Lenovo-iSCSI/ : SUCCESS in 1:05:12\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/850836-5/Lenovo-FC/ : SUCCESS in 1:03:09","accounts_in_message":[],"_revision_number":5},{"id":"4e4ac1d346285ad2051200221ba077ea8d1a28f9","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-08-11 07:16:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"12df7ead86802fbfc46649d0e81aaf88d1ef3bfa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-11 15:51:48.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/36/850836/5/check/DellEMC_Unity_FC/7616a2b/DellEMC_Unity_FC/1433 : FAILURE in 16m 59s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/36/850836/5/check/DellEMC_Unity_iSCSI/e80441c/DellEMC_Unity_iSCSI/1315 : FAILURE in 2h 13m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"1448daff63bbafb6c45e0d11d18ee5d0cdbde658","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-08-12 12:13:40.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/36/850836/5/20220812174018/NEC-FC : SUCCESS in 3h 28m 46s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/36/850836/5/20220812174018/NEC-ISCSI : SUCCESS in 3h 33m 19s","accounts_in_message":[],"_revision_number":5},{"id":"ef4dd9b95742cc12091a69833b79e2ed2e3d0636","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-08-19 14:29:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9ae31fdc0b80e9c96c383e532fa1b1b0057ac89f","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-08-19 15:53:26.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4c80d5b54a252fc84fb921902f7e359e69a00860","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-08-22 11:50:06.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7e91b4be821da7381cc5dea22c550b2d01cd6076","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-08-22 11:51:28.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"6fdcd64c96a1e21064767dd3275a2c64f227274c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-22 12:08:22.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661169012380 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"802aa7564d0e9fcfaf4934125c55b7acc7febf3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 12:09:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bb95b92851c4ba38fdf0970807b7508\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/09d78f875b714ffabbba045906bd2077 : SUCCESS in 13m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9bb3b89f41c84646bcba7b46c709728c : SUCCESS in 14m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9c7e708768e5b1ab73e7e72d663ecfc28c87ae8f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-22 12:35:52.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/36/850836/6/check/dsvm-tempest-cisco-zonemanager-job/210ffbc/ : POST_FAILURE in 13m 39s","accounts_in_message":[],"_revision_number":6},{"id":"c78c0a928c86f89d9654f6c398c253e382a6cfde","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-22 13:02:46.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850836/6/iscsi/ : SUCCESS in 1h 12m 06s\n- cinder-smb http://cloudbase-ci.com/cinder/850836/6/smb/ : SUCCESS in 1h 06m 17s","accounts_in_message":[],"_revision_number":6},{"id":"3638c35457f04b90a7c5d8518aead63274398234","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-22 13:19:49.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850836 : SUCCESS in 1h 28m 08s","accounts_in_message":[],"_revision_number":6},{"id":"f74dfa9138d7bf7e9db6ee1352bf1499c47a5622","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-22 13:28:35.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/36/850836/6/check/fujitsu-eternus-dx-iscsi/7983620 : FAILURE in 1h 37m 51s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/36/850836/6/check/fujitsu-eternus-dx-fc/0e6b9cf : SUCCESS in 1h 27m 12s","accounts_in_message":[],"_revision_number":6},{"id":"12a97ed6b75fda39dbc9026395d50b2ec96084e2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-22 13:42:59.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/36/850836/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5a6ad2e/ : SUCCESS in 1h 42m 47s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/36/850836/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/cd669bb/ : SUCCESS in 1h 51m 48s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/36/850836/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/87ae63d/ : SUCCESS in 1h 27m 27s","accounts_in_message":[],"_revision_number":6},{"id":"0fef3545077746b19c0cb358d2ae5751f4c8b0fe","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-22 13:55:48.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/36/850836/6/check/tempest-dsvm-full-focal-py3/cc76e74/ : SUCCESS in 1h 26m 32s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/36/850836/6/check/devstack-plugin-ceph-dsvm-focal-py3/fe0a4cb/ : SUCCESS in 2h 01m 01s","accounts_in_message":[],"_revision_number":6},{"id":"b3a7234659cc69d0497fb25c49e939f9a8aff376","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-22 14:07:40.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/36/850836/6/check/DellEMC_Unity_FC/a2f1f4e/DellEMC_Unity_FC/1503 : FAILURE in 14m 04s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/36/850836/6/check/DellEMC_Unity_iSCSI/f574563/DellEMC_Unity_iSCSI/1399 : SUCCESS in 1h 47m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"15be84df2583443d332aa99c16396490525eecf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 14:18:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98fd5f9f20954413bd5ebbb0210c078a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38b724994e994bf7b6510ea9ad8bd0ff : SUCCESS in 8m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8cc1e620b774b1dafd22d69258091ec : SUCCESS in 14m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b83d1080a1f44d60a93caaababbbe99b : SUCCESS in 14m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c1331ce3009441dabe91a01373ed3eb : SUCCESS in 18m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64a9d3324daf469f8b95ad5d6f7e0e23 : SUCCESS in 13m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/72fff4d34d0d4532ad827c8723231da8 : SUCCESS in 1h 12m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8e31c923c259498585f24c14e084e34c : SUCCESS in 1h 27m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/158da4cacffb40d2832aa9c291b0d36d : SUCCESS in 1h 05m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e52bc6ff6b5b4eeaabbeb3490299a386 : SUCCESS in 8m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3e7dd2f92ecf48e78ff14f54eec22bb8 : SUCCESS in 20m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ca4fe332bdf24348b75ebfea971365b5 : SUCCESS in 7m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4d7e59ef8be443688096b51bd23ac447 : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f51ac028f5bd4a59bd1dfdf3322e6634 : SUCCESS in 8m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7c59ca8a332942e6bbdce841a4c72767 : SUCCESS in 9m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0a11088b2e1341f4ac77181de6fb71c1 : SUCCESS in 47m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5615df50a8404c889312ed74c44ac9f4 : FAILURE in 8m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f7db26e5d4d0464b883c925d5f2f5ff3 : SUCCESS in 1h 29m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a64f22d62c84449fbb6415887baaf331 : FAILURE in 1h 45m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1c41c09f407d4d7cba1f08b5792cb85f : SUCCESS in 1h 45m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0137e7a01ca54d3caa8bfba7c463b368 : SUCCESS in 1h 24m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/88929490397d4335bc675ab5d3e3360c : SUCCESS in 1h 31m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0f856c44bbe843409770393f720d9776 : SUCCESS in 49m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2b386e5c70eb4c099e20d834bf566cf3 : SUCCESS in 1h 35m 09s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5e5f13d4e0a249faabfe7abdf6964f59 : SUCCESS in 2h 20m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/decc4de776604b39b0a67f2899772a98 : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":6},{"id":"d5946bc92788183bd3029e8def6728a8e5ed9258","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-22 14:52:29.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/36/850836/6/check/iscsi-driver-alletra-6k/8410be2/8410be2.html : TIMED_OUT in 3h 01m 19s","accounts_in_message":[],"_revision_number":6},{"id":"f681a9e1e95ef2a64d09b06e78c929d6e45338b7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-22 15:02:20.000000000","message":"Patch Set 6:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/22_08_2022_14_30_05\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"cd33ad8539f7f5f200c96f92cabfb926a51888d7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-22 16:19:19.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/36/850836/6/check/DellEMC_PowerFlex/75850bc/DellEMC_PowerFlex/1188 : ABORTED in 2h 00m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"dd073691248c77406ec4e27d89b78cf187b6efb8","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-08-22 16:33:52.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/850836-6/HPMSA-iSCSI/ : SUCCESS in 54:04.71\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/850836-6/HPMSA-FC/ : SUCCESS in 51:11.71","accounts_in_message":[],"_revision_number":6},{"id":"96424be6d422de0c0a94ae9e74a5d70f1f0060c0","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-08-22 16:54:41.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/850836-6/Lenovo-iSCSI/ : SUCCESS in 1:04:35\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/850836-6/Lenovo-FC/ : SUCCESS in 1:02:02","accounts_in_message":[],"_revision_number":6},{"id":"ff370389a24a9835a584e47547f493b9c9b04529","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-08-22 17:08:07.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/850836-6/PVME-iSCSI/ : SUCCESS in 1:09:22\n- PVME-FC https://os-logs.seagate.com/dell-ci/850836-6/PVME-FC/ : SUCCESS in 1:09:51","accounts_in_message":[],"_revision_number":6},{"id":"8fc3447c48c97df524bb853b37de4e95557231e1","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-08-22 17:09:35.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/850836-6/Seagate-iSCSI/ : SUCCESS in 1:10:55\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/850836-6/Seagate-FC/ : SUCCESS in 1:09:14","accounts_in_message":[],"_revision_number":6},{"id":"5ab43746defcff99c98f70423529bc0e605361c0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-22 17:44:06.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/36/850836/6/check/DellEMC_Unity_FC/006aebb/DellEMC_Unity_FC/1504 : FAILURE in 14m 32s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/36/850836/6/check/DellEMC_Unity_iSCSI/b930c78/DellEMC_Unity_iSCSI/1400 : SUCCESS in 1h 43m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"c31f45392f146b13498390d8f0c8dfb8588ce722","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-22 18:01:06.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/36/850836/6/check/DellEMC_PowerStore_FC/7fe9910/DellEMC_PowerStore_FC/757 : SUCCESS in 1h 15m 20s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/36/850836/6/check/DellEMC_PowerStore_NFS/456d7fd/DellEMC_PowerStore_NFS/None : NOT_REGISTERED\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/36/850836/6/check/DellEMC_PowerStore_NVMeTCP/bd902df/DellEMC_PowerStore_NVMeTCP/172 : SUCCESS in 2h 06m 01s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"2ffd83cf624d7a3e15696f00d420b56a8771416e","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-08-22 18:17:37.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/36/850836/6/check/DellEMC_SC_FC/c512fbb/DellEMC_SC_FC/945 : SUCCESS in 52m 58s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/36/850836/6/check/DellEMC_SC_iSCSI/bf0ca8a/DellEMC_SC_iSCSI/963 : SUCCESS in 2h 19m 48s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":6},{"id":"d8e5e7e9f723768cbb473455915a008ecc42c8c3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-08-23 00:04:57.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/36/850836/6/20220823015550/NEC-FC : SUCCESS in 7h 06m 09s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/36/850836/6/20220823015550/NEC-ISCSI : SUCCESS in 7h 09m 04s","accounts_in_message":[],"_revision_number":6},{"id":"84c39af395b53b0238e49e61eeabf8ae12876135","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-26 14:13:23.000000000","message":"Patch Set 6:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/476","accounts_in_message":[],"_revision_number":6},{"id":"b157b585076f725873e871f712f8db72e8b11c03","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-08-29 01:38:52.000000000","message":"Patch Set 6:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":6},{"id":"209882562304e2b76cdd32dd13e9eddb0f4528e8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-08-29 01:40:38.000000000","message":"Patch Set 2:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":2},{"id":"8acf2d4aafa2f95c3b85b5cf47f4b70a6562e634","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-08-29 01:40:42.000000000","message":"Patch Set 4:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":4},{"id":"27e1dc53bbcfa7d003218ff2ee2097f8e36c90fd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-08-29 01:40:45.000000000","message":"Patch Set 3:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":3},{"id":"e187dfcc13faf569bbd8fbbdf77b3550c3ca478f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-08-29 01:42:44.000000000","message":"Patch Set 5:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":5},{"id":"2043e87012f4e116414c41e2c1a22c3f35630609","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-08-31 15:32:21.000000000","message":"Patch Set 6:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/36/850836/6/20220831174833/NEC-V-FC : FAILURE in 5h 29m 08s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/36/850836/6/20220831174833/NEC-V-ISCSI : FAILURE in 6h 43m 45s","accounts_in_message":[],"_revision_number":6},{"id":"3729dcde20729636710114c9d6528f75517cb436","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-09-01 14:37:26.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ff04f230c85d90e32167367010a4574407416ec4","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-09-02 04:51:39.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"acf4393068cdd308ce3b6fe480b2eee97d799550","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-09-06 12:08:35.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4a87e1dadbe91bef26c2db8a1b7a989f39553e6d","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-09-06 12:09:53.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e81ee456df7d94461adda59d06cddafa45abb907","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-06 12:15:29.000000000","message":"Patch Set 7:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/559","accounts_in_message":[],"_revision_number":7},{"id":"793833ee56ab41100e0adffd2ee7d3d725d5fb27","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-09-06 12:35:13.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/36/850836/7/check/iscsi-driver-alletra-6k/dcd6e17/dcd6e17.html : FAILURE in 25m 29s","accounts_in_message":[],"_revision_number":7},{"id":"f12f94dbc06ac4f4acc4086981a0917d6aba1389","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 12:43:04.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d5e5e8a9626424689799bf545661a21\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c42297fba37e4497aa9cc304a8316241 : FAILURE in 18m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8b77df2feb7f4a7e885dd6be6a7d1fa9 : FAILURE in 31m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a511f5e34964ba7762197ffe1da869996359f053","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-09-06 13:10:31.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850836/7/iscsi/ : SUCCESS in 1h 01m 14s\n- cinder-smb http://cloudbase-ci.com/cinder/850836/7/smb/ : SUCCESS in 55m 45s","accounts_in_message":[],"_revision_number":7},{"id":"0e6321ae325ea69c30f002e074241730ff602938","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-09-06 13:29:58.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/36/850836/7/check/dsvm-tempest-cisco-zonemanager-job/e866825/ : SUCCESS in 1h 05m 18s","accounts_in_message":[],"_revision_number":7},{"id":"32520880e02f16ba13065e016728ed69e6556faf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-06 13:35:28.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/850836 : SUCCESS in 1h 25m 11s","accounts_in_message":[],"_revision_number":7},{"id":"97fb1f2d83843d067a109ece83bb7a1685ddab7b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-06 13:38:17.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850836 : FAILURE in 1h 27m 50s","accounts_in_message":[],"_revision_number":7},{"id":"6e9e928f48ad46e997014355af704bf655105a6f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-06 14:08:18.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/36/850836/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/07fed1a/ : SUCCESS in 1h 34m 48s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/36/850836/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/64f0e27/ : SUCCESS in 1h 58m 47s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/36/850836/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/1edb314/ : SUCCESS in 1h 32m 56s","accounts_in_message":[],"_revision_number":7},{"id":"a48490b095f635da9d719087f1009db2ff83f0cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 14:23:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2c0feb9e6ba4bbbb7a6f7511a998d86\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3782a3e558f46af82413129c9654708 : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/412c5085c6a7429bb7ee2114bb0d3dc4 : SUCCESS in 11m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8620a89136af407b9756d91b249f064d : SUCCESS in 11m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a08cceb6b3f446849ff5cbc6d23fce30 : FAILURE in 26m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b01d8a026ece46318e9058c05c179266 : SUCCESS in 13m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/21fa08bcc6004c2ca3a2a29b557d0e2e : SUCCESS in 1h 01m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/106df20d583a4703a0aff6e55a9f068b : SUCCESS in 1h 23m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9ed376bf1fd44c880f31738981092e5 : SUCCESS in 1h 09m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78234c50c1144894ade040438ba8f874 : SUCCESS in 7m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d5e2c896775b45deb28fd742468c69d1 : SUCCESS in 14m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/88dc34f3f17e4b07aaa896b9771d0850 : SUCCESS in 6m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d14e004561b544a0a9faaa8d7622c1ee : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b0d46f16143c4f169f37cfbe27ce4d05 : SUCCESS in 6m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/be0b731a21994141a73f132589c517a5 : SUCCESS in 6m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4ae43433f749470e9626d8dc1f7b7886 : SUCCESS in 44m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/305d0fcf13fe4762a3fe167f51fabd08 : FAILURE in 7m 11s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3ca0370d5b3b4829909721796bdcf23a : SUCCESS in 1h 31m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e48f14314a2e434aaa10e93df9c282f0 : FAILURE in 1h 37m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3a2f8cefe9fc43aea1b8ed33c8d60620 : SUCCESS in 1h 41m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b3943ef3c3754b5e899773ce311a39d0 : SUCCESS in 1h 01m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e39326078a88436bbe8905c5daa2c0b4 : SUCCESS in 1h 26m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ea82415debdc44099bf173c2bcaab0ec : SUCCESS in 38m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/efac532602c64c089f8cc8ad36fd77dc : SUCCESS in 1h 16m 51s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/79122b7c4dfc4c78a55b743c3c07987e : SUCCESS in 2h 13m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df7d95d3603240fca37854ba33329210 : SUCCESS in 52m 23s","accounts_in_message":[],"_revision_number":7},{"id":"9fa624cb56c0638cdf5b30212d305e51db666a10","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-09-06 14:44:58.000000000","message":"Patch Set 7:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/06_09_2022_14_23_56\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"3f90284ba92c9eaedda06f735c40f11774f65bef","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-06 15:34:56.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/850836/7/upstream-check/cinder-tempest-plugin-ontap-fcp/48b3d6c/ : FAILURE in 1h 10m 13s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/850836/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/8e49dda/ : FAILURE in 31m 55s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/850836/7/upstream-check/cinder-tempest-plugin-ontap-nfs/d659c36/ : SUCCESS in 1h 03m 23s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/850836/7/upstream-check/cinder-tempest-plugin-solidfire-iscsi/04ffe2b/ : SUCCESS in 50m 44s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"3cb85334f8f3b7f5fc4886c57d7f276bf9d39d0a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-06 15:36:52.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-36-850836-7/ns5_nfs : SUCCESS in 3h 27m 49s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-36-850836-7/ns5_iscsi : SUCCESS in 3h 27m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"fa3388838df50a4912537babb5a4c22d6f8a86b5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-09-06 17:22:52.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/36/850836/7/check/DellEMC_PowerFlex/b2bc590/DellEMC_PowerFlex/1267 : ABORTED in 2h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"e86b53bd42894a84f7664e3da119235431f76282","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-09-06 18:25:42.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/850836-7/HPMSA-iSCSI/ : SUCCESS in 59:09.05\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/850836-7/HPMSA-FC/ : SUCCESS in 57:56.16","accounts_in_message":[],"_revision_number":7},{"id":"bdd5d03a1ad07b8b285a63a293e61348d5659455","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-06 18:55:43.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/36/850836/7/check/DellEMC_Unity_FC/8a895af/DellEMC_Unity_FC/1696 : SUCCESS in 1h 06m 36s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/36/850836/7/check/DellEMC_Unity_iSCSI/322aac2/DellEMC_Unity_iSCSI/1557 : SUCCESS in 2h 06m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"7c9ac7927c7e41c532130f79624ec038c589d0f4","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-09-06 19:11:38.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/850836-7/Seagate-iSCSI/ : SUCCESS in 1:11:03\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/850836-7/Seagate-FC/ : SUCCESS in 1:13:38","accounts_in_message":[],"_revision_number":7},{"id":"74d7fbc30cfbc2cf341be1b3ecef41fa5ee31cdb","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-09-06 19:13:17.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/850836-7/Lenovo-iSCSI/ : SUCCESS in 1:10:07\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/850836-7/Lenovo-FC/ : SUCCESS in 1:06:59","accounts_in_message":[],"_revision_number":7},{"id":"979a2d915038f7ff10b4ecd2e5772d65fb86067c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-09-06 19:22:51.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/36/850836/7/check/DellEMC_PowerStore_iSCSI/3f55636/DellEMC_PowerStore_iSCSI/709 : SUCCESS in 2h 00m 29s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/36/850836/7/check/DellEMC_PowerStore_FC/5298966/DellEMC_PowerStore_FC/863 : SUCCESS in 1h 15m 07s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/36/850836/7/check/DellEMC_PowerStore_NFS/6b3c433/DellEMC_PowerStore_NFS/398 : ABORTED in 3h 00m 01s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/36/850836/7/check/DellEMC_PowerStore_NVMeTCP/a46a6ab/DellEMC_PowerStore_NVMeTCP/273 : SUCCESS in 1h 41m 48s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"5ea86d77c5ec8d4bfc1c3f0e75029654b182de44","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-09-07 11:39:30.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"dcca544a7a6cbfdf18f68b94771e0e244863a6ae","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-07 12:58:03.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850836 : FAILURE in 1h 09m 24s","accounts_in_message":[],"_revision_number":7},{"id":"3548facbe676190a68977a593793cf78b67a62c5","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-09-08 13:09:34.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/36/850836/7/ : FAILURE in 1h 18m 36s","accounts_in_message":[],"_revision_number":7},{"id":"3645b38d932c2e9e12935db1f6254931341c7746","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-09-09 12:47:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"390ab6c0f8404786f798c1c31f8637df5f8956d0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-09 15:24:06.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850836 : FAILURE in 1h 26m 39s","accounts_in_message":[],"_revision_number":7},{"id":"9830a141a6e58dfaee1e1bca5cc5d8cb30e3d939","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-09-11 21:14:18.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/36/850836/7/20220911225729/NEC-FC : SUCCESS in 7h 07m 26s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/36/850836/7/20220911225729/NEC-ISCSI : SUCCESS in 7h 16m 46s","accounts_in_message":[],"_revision_number":7},{"id":"6c8f470ef3ce7fe75a2e533d1f3c9c884add0bb1","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-09-12 05:56:02.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6ee00091fa4f1fdb1b4847bbfb05123f7e9d199e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-12 07:30:29.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850836 : SUCCESS in 1h 32m 43s","accounts_in_message":[],"_revision_number":7},{"id":"81549e9f332a5f9289277ee117b77fa3ffe7f8f2","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-09-12 16:18:50.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d4bb0e34ab8266aca4290f7d01b487023e086bc4","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-09-13 03:42:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8ed8efe1207546cdec114ac7ef365c3a1b6ce9f6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-09-13 16:45:22.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a826896e21214ae092343836d98fd76f : SUCCESS in 37m 50s","accounts_in_message":[],"_revision_number":7},{"id":"840ebd6ca6ed66a2cf57888ce762ff36fddb0b20","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-09-14 13:38:11.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"7b54404e4a7d71c37743f59aa011ff8a09d60279","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 13:39:36.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"0afce5a3099f17579848ef50ba194f9b864279ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 15:46:04.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc7046925dd34f98bbf54a60fc6ad740\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3f669c7a91c44c8b5c6e820791b42e6 : SUCCESS in 10m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9ba32a8d0604792b59cf8881435db81 : SUCCESS in 12m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cc9dd0d46c84875a64629e56b43a090 : SUCCESS in 12m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3362748f679d46ee92832bcf4a559cc8 : SUCCESS in 13m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/6d7f285560874afb8becee175bc1109a : SUCCESS in 1h 03m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2fccd64e1c944f30b2401bde0f8f890c : SUCCESS in 1h 57m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9ef9b1cf1ab460f83889d64ecfbd212 : SUCCESS in 1h 25m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c05a1114da074a76a19c9f8977ceeec5 : SUCCESS in 9m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f49b0483f91e4645aa415db8c7525430 : SUCCESS in 1h 19m 34s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/78586ffa6bc642898f273af4fcedd38d : FAILURE in 1h 24m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce7e41742fa34e0f96a1ab62f85a0790 : SUCCESS in 1h 02m 07s","accounts_in_message":[],"_revision_number":7},{"id":"71e13f68b995a775d7021722438c9fd8537fe01d","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-09-15 03:55:24.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c6f59ab13e47580e980328f7423bbe112f750a7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 05:00:17.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7a52968766c4b579c075621a2af21ae\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4c54de7add44b5cb78e8b1012d27e33 : SUCCESS in 20m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/417e7d47757046a9a371a5320cfe8255 : TIMED_OUT in 1h 00m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"23b9db187d4f4a26849dcfec0bc5664b1a57a3aa","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-09-15 05:01:33.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/7491ba9feccf419d92d6ca6f4777098c : SUCCESS in 1h 05m 20s","accounts_in_message":[],"_revision_number":7},{"id":"28e34872fe527b28e0ac7de4ef4903331e1acc8c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-09-15 05:10:34.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/36/850836/7/check/dsvm-tempest-cisco-zonemanager-job/6275cfe/ : FAILURE in 1h 07m 54s","accounts_in_message":[],"_revision_number":7},{"id":"090844e8e7716ae009cded5386cf5127ee0435f5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-09-15 05:11:50.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850836/7/iscsi/ : SUCCESS in 1h 15m 58s\n- cinder-smb http://cloudbase-ci.com/cinder/850836/7/smb/ : SUCCESS in 58m 39s","accounts_in_message":[],"_revision_number":7},{"id":"07c74139ccb5e532f87c102d907849cfd9e941b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 06:24:13.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7ffd892a99741268816880420716869\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ded1f969ec84b5abfa527d2eb42e195 : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/098dac2253dd48c382b8dd63d523ede7 : FAILURE in 11m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1845a001858f4a85be1d8a0393fabadf : SUCCESS in 10m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2c0a33f894e4e269f863533aa39dd44 : SUCCESS in 9m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a600a07205284b3f94c62b4e51fe2d0b : SUCCESS in 13m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/333971bc2f2349cd9ee1d3c5c6fa2f57 : SUCCESS in 44m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c5abdf50afcb47e690d5076ebd42dcdc : SUCCESS in 1h 26m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82454a89687643eca6d9c92abbf7a38d : SUCCESS in 1h 08m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84084a4b40e24a1bb57e05e55479cb7c : SUCCESS in 7m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/13a4fd00bc3c47e292b5a26acd77334f : SUCCESS in 12m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4a5afb4784c7475e944df6bdf2cc0d9d : SUCCESS in 6m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2fc0515cd2bc4079816916fc98f8101d : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6b44196f21ce41c082fe59eec496f6b1 : SUCCESS in 6m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/225671b1f7b547939c6f7a5bee9d7de0 : SUCCESS in 6m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0dfa0b2c89224b309fa78119df72461d : SUCCESS in 30m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8c4649d889cb4a9d99aac0295e572770 : FAILURE in 6m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eab48f6d25f14c8eab27d8709535a36e : SUCCESS in 1h 04m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/01960948f0244b9da4e89decfef83d26 : FAILURE in 1h 43m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8fd115a0bbe046868c09c690533a72e2 : SUCCESS in 1h 54m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/da941a347e54483fb87709f84cde934b : SUCCESS in 1h 13m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b4eb501cd5d14e378918d77e59c05e3b : SUCCESS in 1h 39m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e989572f72a7468cbcb3f30dead3e50a : SUCCESS in 48m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/83d1ecebef314600b2d97907089df0e6 : SUCCESS in 1h 52m 22s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d71d10cb9be64ab0a2360db09e58b853 : SUCCESS in 2h 21m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fcbb32573b841c8a1ec2aebe47453d3 : SUCCESS in 1h 00m 24s","accounts_in_message":[],"_revision_number":7},{"id":"252df3bc4f58053c7830bc01558c9bc694c20103","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-09-15 08:38:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"94038c86bd21646b46d45c69356485e2a842b737","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 09:06:35.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d67696bb766146efa1afa2000ce51705\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8aad7f72f9b4110910601e3cdb75f8d : SUCCESS in 26m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/331df710ae764453bc588f50d9cbf626 : SUCCESS in 15m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7351511d6e13a7107c852d7847fe244b10a6c584","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-09-15 09:40:22.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/027964ff96c1482d9942b647c2d1445c : SUCCESS in 1h 01m 10s","accounts_in_message":[],"_revision_number":7},{"id":"33383165fc4c0b078234e5606332535b3e4547bb","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-09-15 09:44:02.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850836/7/iscsi/ : SUCCESS in 1h 05m 07s\n- cinder-smb http://cloudbase-ci.com/cinder/850836/7/smb/ : SUCCESS in 53m 12s","accounts_in_message":[],"_revision_number":7},{"id":"0bb95fd22069e13b3374611b0f9b4824739882c1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-09-15 10:02:55.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/36/850836/7/check/dsvm-tempest-cisco-zonemanager-job/3e08604/ : FAILURE in 1h 08m 14s","accounts_in_message":[],"_revision_number":7},{"id":"7403e40d4c56d5cd394b655abaccda558cff9011","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 10:40:18.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/611a61b013c040b4b7e141f1fa43b7b9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01a69c52d9524f439683e3a39fd39aa3 : SUCCESS in 8m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d0c2115f6e64c68bfd8bfbc66102180 : SUCCESS in 10m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bd918e5692444c5982e33ea9b7fa200 : SUCCESS in 12m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6582bcf988241be8cb925ed2d4655f3 : SUCCESS in 11m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2b6314cc89b4aecaf75350fd3389fc9 : SUCCESS in 11m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/d02adaa2d1bf4c768feeb7b80252f8ca : FAILURE in 1h 02m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b162f5805b37427b8a1a8fc291838388 : SUCCESS in 1h 26m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92d0211188294740b5b72e6bdccb02cf : SUCCESS in 1h 19m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a385dddb25584d85a8ade63b2c631807 : SUCCESS in 10m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ba1d74d33a2b4d60adf7aeb4a471291c : SUCCESS in 14m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/14f6cb4697534943b2863c14d2bb9ecd : SUCCESS in 6m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/af54d50ecd1046c9a745a852da0e6a8d : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d3948a6693a34ecba97ebdf972b72f4a : SUCCESS in 7m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a853ffaf6b1443be88844fe6a5fb6bbc : SUCCESS in 7m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d62907b25dc1442088da143324770bda : SUCCESS in 42m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/04991055ddad44ba83bf9442d5db7072 : FAILURE in 7m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/54cf4ec49b2d4e5baf8830247b85d05e : SUCCESS in 1h 20m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/053f625cc90041e1ab54d846f997d9b5 : FAILURE in 1h 31m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/66afbebaf21841fe99d2cd5cdce60d41 : SUCCESS in 1h 41m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3a629cedd771434aa80d972a23c6bff0 : SUCCESS in 1h 17m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0019a6dbd2184db28562937839326eee : SUCCESS in 1h 22m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/32bce1cad3d44240a54a95ef809736d9 : SUCCESS in 44m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3e9fc9e634c14d29a14c8abdb9492bb2 : SUCCESS in 1h 25m 46s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2744e17803a44162b0396115b4b0a967 : SUCCESS in 1h 59m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5dc59ec24454fa6a5f4cc242907a28a : SUCCESS in 53m 16s","accounts_in_message":[],"_revision_number":7},{"id":"71da5a8321cd94d11dfeb89d9e3c73482ef5da2f","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-09-15 11:43:30.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"12c201bcba82826bc4cdc5aa47ff55cb0d998687","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 12:08:53.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff0566c917354ab6819a300cfb8e8737\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e4ba67ad6d84f678fc1ce3d9065f3b0 : SUCCESS in 21m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70d56d0141ba4b42bf69841ea52515b8 : SUCCESS in 23m 30s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"55d6c46b01f832af2fa93a46ed3f96f06221734a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-09-15 12:47:57.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850836/7/iscsi/ : SUCCESS in 1h 03m 59s\n- cinder-smb http://cloudbase-ci.com/cinder/850836/7/smb/ : SUCCESS in 58m 37s","accounts_in_message":[],"_revision_number":7},{"id":"b49f61e686084caddb2ac4039c8a192cbad95c76","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-09-15 13:06:41.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/36/850836/7/check/dsvm-tempest-cisco-zonemanager-job/f1aea47/ : SUCCESS in 1h 08m 56s","accounts_in_message":[],"_revision_number":7},{"id":"746f79719f342bec297bb4f1cdf8cae40ebe182b","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-09-15 13:10:20.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/25223ef1e7c8400daa3c5f4d94f5d5cf : SUCCESS in 1h 26m 00s","accounts_in_message":[],"_revision_number":7},{"id":"15e0e66aa80f24285ad9a94f71c1b4fbba31d492","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 14:15:04.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54e4b937e1fb471b85692021f5f51e7f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d71dc817fe748588d8eac4de57212b1 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e677c6c50c984545bb09f18d600c0cf3 : SUCCESS in 11m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f8f329f09a74805ab4f3d5fd0efc160 : SUCCESS in 11m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7131d068919e4a2992cd16dfdf193f9b : SUCCESS in 11m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c12ea78281ed40c0abb2b6f5987563f0 : SUCCESS in 12m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/9022bcfceae64cfc90467f22dcf598e4 : SUCCESS in 57m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f411669899d1440986e7c2ab17ed3638 : SUCCESS in 1h 14m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e706ebef143844ddaedf78b8a977751b : SUCCESS in 1h 13m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11d5a62e559f47aab38f712be00caa77 : SUCCESS in 8m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7f499a80eb6f417e97e1715895a4f8dc : SUCCESS in 18m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3edbc823855f4d27a373880792f9cacb : SUCCESS in 5m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/974b909847594bf2b34ea6248bb6e830 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1e9ee391e4ca44fe89ce1f29411a4184 : SUCCESS in 6m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5aca375898bf4dee8afc23da989242f6 : SUCCESS in 6m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3ee9d24771594957a25e19e11b73d6a7 : SUCCESS in 47m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/819e8d8f483c4a769df8205e662c19ff : FAILURE in 9m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/805ca1b547dd496fb0b6eb6f7b5c2953 : SUCCESS in 1h 33m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8644936a51ea47a5a141566057135403 : FAILURE in 1h 41m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ea8cfb6007144d4aae60bb9db2e344aa : SUCCESS in 1h 49m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/583a903fe9c044b1a6d1483a231540ed : SUCCESS in 1h 20m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6c52d0afa08248528b2ca6a4b7e91472 : SUCCESS in 1h 34m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1d9e32aa75734b0b8a6d5960b730a4e2 : SUCCESS in 43m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d3caea7900a840fb862336f492d4d623 : SUCCESS in 1h 27m 08s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/da450b4d2216493abcb7c6586ea6ef37 : SUCCESS in 2h 26m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7576255609d40f8a0422d4356cb1ef0 : SUCCESS in 1h 00m 04s","accounts_in_message":[],"_revision_number":7},{"id":"9120d5fcb188f562139c0c2ce73fdb4cf04a82d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 14:16:01.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"970371ced70ae61294aa0fbdd3dfd056e9bf09ed","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-09-15 14:22:43.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/15_09_2022_14_15_13\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"639fb88c42743fe1ab955498c66230cdbfe61bd1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-15 15:14:05.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/850836/7/upstream-check/cinder-tempest-plugin-ontap-fcp/7c860ce/ : SUCCESS in 58m 08s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/850836/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/583dd18/ : SUCCESS in 40m 54s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/850836/7/upstream-check/cinder-tempest-plugin-ontap-nfs/fe78f04/ : SUCCESS in 44m 20s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/850836/7/upstream-check/cinder-tempest-plugin-solidfire-iscsi/3869752/ : SUCCESS in 53m 51s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"aed24317ec16c3574c06fc6eb2dd4e1cab907104","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 16:48:00.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/680233c8f3fd40f3b08a20d3fd5deaa4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f9eff74c37048959e393b8b70cb80cd : SUCCESS in 8m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5377f083092c4c7da12ab9a655f820ec : SUCCESS in 11m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee8598fce4d74c46b4d2c174b7341e55 : SUCCESS in 12m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aad14c147f94a248a7dee657adbba9e : SUCCESS in 13m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/b8a25821b37d4cf991b97ed2395d22fe : SUCCESS in 59m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/79f4bc4029e64262ad49e00e86bdc6af : SUCCESS in 1h 30m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/461284448b984e6eaf65d786b98400fd : SUCCESS in 1h 11m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66554f14ee0f4427836fdb8397004021 : SUCCESS in 9m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b5fd404eaf574581a32f466db7cee94d : SUCCESS in 1h 20m 48s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/93711554b94446b692aae859491261b7 : SUCCESS in 1h 11m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6476fdd991341cb89e7dd4a8c7d6587 : SUCCESS in 47m 15s","accounts_in_message":[],"_revision_number":7},{"id":"5f8e0da75d57f4300dbc50d0b1896dbda6af2a2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 16:48:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"ddae292d2198c9df874cbdb28b4009f01cace551","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 16:49:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c5071f62b014dd4b563d61d3e7ae77e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c2ba15de1f54de184aed775b4d53e50 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27d382c945094175b84a9d3c8de8daac : SUCCESS in 53s","accounts_in_message":[],"_revision_number":7},{"id":"4dd1ba810c9c79aea0021fab894c23106c807df2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-09-15 17:58:08.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/36/850836/7/check/DellEMC_PowerFlex/d4d5d1d/DellEMC_PowerFlex/1326 : FAILURE in 2h 00m 40s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"2d7cda0ef5c580687ed5e56c0d95c05df7fd1c2e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-09-15 18:31:29.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/36/850836/7/check/DellEMC_PowerStore_iSCSI/2e35df3/DellEMC_PowerStore_iSCSI/769 : FAILURE in 1h 02m 40s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/36/850836/7/check/DellEMC_PowerStore_FC/81ff1a4/DellEMC_PowerStore_FC/923 : SUCCESS in 1h 20m 49s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/36/850836/7/check/DellEMC_PowerStore_NFS/7caeaff/DellEMC_PowerStore_NFS/458 : SUCCESS in 2h 00m 42s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/36/850836/7/check/DellEMC_PowerStore_NVMeTCP/56c7388/DellEMC_PowerStore_NVMeTCP/333 : FAILURE in 1h 34m 47s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"f0adcfc34747c55fe0e407bc8a680d4bd80281c3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-15 19:55:45.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/36/850836/7/check/DellEMC_Unity_FC/0701fe7/DellEMC_Unity_FC/1793 : FAILURE in 1h 19m 22s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/36/850836/7/check/DellEMC_Unity_iSCSI/3128188/DellEMC_Unity_iSCSI/1643 : FAILURE in 1h 42m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"31e8a29161c0b41b291aa2bd5c884ab97dceffb0","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-09-16 04:06:25.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2901add2e528737b9d295cbf85094e71808a5950","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-09-17 02:17:44.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/dc18652fd3a94d0c99719c3578b1cea5 : TIMED_OUT in 3h 04m 51s","accounts_in_message":[],"_revision_number":7},{"id":"3b52c5c0c8dd6f1b602d78c54c0f625a72bfa290","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-09-17 20:49:57.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/850836-7/Lenovo-iSCSI/ : SUCCESS in 1:06:16\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/850836-7/Lenovo-FC/ : SUCCESS in 1:04:37","accounts_in_message":[],"_revision_number":7},{"id":"2e1035ec62a73d30d19fe6ef67481beb091fca9a","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-09-18 07:54:37.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/850836-7/HPMSA-iSCSI/ : SUCCESS in 56:23.36\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/850836-7/HPMSA-FC/ : SUCCESS in 53:39.71","accounts_in_message":[],"_revision_number":7},{"id":"705f886dfb23ce02a90d2300f163de9032e8b1a2","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-09-19 12:02:56.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/850836-7/Seagate-iSCSI/ : SUCCESS in 1:07:06\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/850836-7/Seagate-FC/ : SUCCESS in 1:09:59","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"c112542cf5471641a47a94c89b9b80dfa14d701d","revisions":{"6c72c7575dc7c8075966c6c7ccdb2fa4c6d29844":{"kind":"REWORK","_number":1,"created":"2022-07-25 07:18:10.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/36/850836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/850836/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/850836/1"}}},"commit":{"parents":[{"commit":"a2deecef891a16703e5410eb7805442961307090","subject":"Merge \"mypy: service.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2deecef891a16703e5410eb7805442961307090"}]}],"author":{"name":"raghavendrat","email":"raghavendra.tilay@hpe.com","date":"2022-07-25 06:59:55.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra.tilay@hpe.com","date":"2022-07-25 07:13:36.000000000","tz":0},"subject":"HPE 3PAR: Umanaged volumes \u0026 snapshots missing from cinder manageable-list","message":"HPE 3PAR: Umanaged volumes \u0026 snapshots missing from cinder manageable-list\n\nAdded below two functions:\nget_manageable_volumes()\nget_manageable_snapshots()\n\nCloses-Bug: #1819903\nChange-Id: Ie178e48958a69ef2b6c329e132de2c772d5c1a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c72c7575dc7c8075966c6c7ccdb2fa4c6d29844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6c72c7575dc7c8075966c6c7ccdb2fa4c6d29844"}]},"branch":"refs/heads/master"},"4db6f619ef36ee5f6ada1fcfbd467c43af5a1942":{"kind":"REWORK","_number":2,"created":"2022-08-03 09:58:40.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/36/850836/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/850836/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/850836/2"}}},"commit":{"parents":[{"commit":"a2deecef891a16703e5410eb7805442961307090","subject":"Merge \"mypy: service.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2deecef891a16703e5410eb7805442961307090"}]}],"author":{"name":"raghavendrat","email":"raghavendra.tilay@hpe.com","date":"2022-07-25 06:59:55.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra.tilay@hpe.com","date":"2022-08-03 09:50:46.000000000","tz":0},"subject":"HPE 3PAR: Fix umanaged volumes \u0026 snapshots missing","message":"HPE 3PAR: Fix umanaged volumes \u0026 snapshots missing\n\nThe umanaged volumes \u0026 snapshots are missing when below commands\nare executed:\ncinder manageable-list\ncinder snapshot-manageable-list\n\nTo address this, added two functions:\nget_manageable_volumes()\nget_manageable_snapshots()\n\nCloses-Bug: #1819903\nChange-Id: Ie178e48958a69ef2b6c329e132de2c772d5c1a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4db6f619ef36ee5f6ada1fcfbd467c43af5a1942"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4db6f619ef36ee5f6ada1fcfbd467c43af5a1942"}]},"branch":"refs/heads/master"},"f44f3209ee1eab4f7b9aaa0b9afabb29299f3aaa":{"kind":"REWORK","_number":3,"created":"2022-08-03 12:05:08.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/36/850836/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/850836/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/850836/3"}}},"commit":{"parents":[{"commit":"a2deecef891a16703e5410eb7805442961307090","subject":"Merge \"mypy: service.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2deecef891a16703e5410eb7805442961307090"}]}],"author":{"name":"raghavendrat","email":"raghavendra.tilay@hpe.com","date":"2022-07-25 06:59:55.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra.tilay@hpe.com","date":"2022-08-03 11:58:14.000000000","tz":0},"subject":"HPE 3PAR: Fix umanaged volumes \u0026 snapshots missing","message":"HPE 3PAR: Fix umanaged volumes \u0026 snapshots missing\n\nThe umanaged volumes \u0026 snapshots are missing when below commands\nare executed:\ncinder manageable-list\ncinder snapshot-manageable-list\n\nTo address this, added two functions:\nget_manageable_volumes()\nget_manageable_snapshots()\n\nCloses-Bug: #1819903\nChange-Id: Ie178e48958a69ef2b6c329e132de2c772d5c1a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f44f3209ee1eab4f7b9aaa0b9afabb29299f3aaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f44f3209ee1eab4f7b9aaa0b9afabb29299f3aaa"}]},"branch":"refs/heads/master"},"3f8a3c870b0a5d1a9296325e6af867f309017c58":{"kind":"REWORK","_number":4,"created":"2022-08-03 12:07:05.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/36/850836/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/850836/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/850836/4"}}},"commit":{"parents":[{"commit":"a2deecef891a16703e5410eb7805442961307090","subject":"Merge \"mypy: service.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2deecef891a16703e5410eb7805442961307090"}]}],"author":{"name":"raghavendrat","email":"raghavendra.tilay@hpe.com","date":"2022-07-25 06:59:55.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra.tilay@hpe.com","date":"2022-08-03 12:02:57.000000000","tz":0},"subject":"HPE 3PAR: Fix umanaged volumes \u0026 snapshots missing","message":"HPE 3PAR: Fix umanaged volumes \u0026 snapshots missing\n\nThe umanaged volumes \u0026 snapshots are missing when below commands\nare executed:\ncinder manageable-list\ncinder snapshot-manageable-list\n\nTo address this, added two functions:\nget_manageable_volumes()\nget_manageable_snapshots()\n\nCloses-Bug: #1819903\nChange-Id: Ie178e48958a69ef2b6c329e132de2c772d5c1a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f8a3c870b0a5d1a9296325e6af867f309017c58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3f8a3c870b0a5d1a9296325e6af867f309017c58"}]},"branch":"refs/heads/master"},"a721158aa1e867ee1cf2eb525b612710cf6731d3":{"kind":"REWORK","_number":5,"created":"2022-08-08 10:36:41.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/36/850836/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/850836/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/850836/5"}}},"commit":{"parents":[{"commit":"a2deecef891a16703e5410eb7805442961307090","subject":"Merge \"mypy: service.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2deecef891a16703e5410eb7805442961307090"}]}],"author":{"name":"raghavendrat","email":"raghavendra.tilay@hpe.com","date":"2022-07-25 06:59:55.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra.tilay@hpe.com","date":"2022-08-08 10:27:24.000000000","tz":0},"subject":"HPE 3PAR: Fix umanaged volumes \u0026 snapshots missing","message":"HPE 3PAR: Fix umanaged volumes \u0026 snapshots missing\n\nThe umanaged volumes \u0026 snapshots are missing when below commands\nare executed:\ncinder manageable-list\ncinder snapshot-manageable-list\n\nTo address this, added two functions:\nget_manageable_volumes()\nget_manageable_snapshots()\n\nCloses-Bug: #1819903\nChange-Id: Ie178e48958a69ef2b6c329e132de2c772d5c1a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a721158aa1e867ee1cf2eb525b612710cf6731d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a721158aa1e867ee1cf2eb525b612710cf6731d3"}]},"branch":"refs/heads/master"},"7d483ad49e9b438b01e1639664b1d77a74edcf45":{"kind":"REWORK","_number":6,"created":"2022-08-22 11:50:06.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/36/850836/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/850836/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/850836/6"}}},"commit":{"parents":[{"commit":"a2deecef891a16703e5410eb7805442961307090","subject":"Merge \"mypy: service.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2deecef891a16703e5410eb7805442961307090"}]}],"author":{"name":"raghavendrat","email":"raghavendra.tilay@hpe.com","date":"2022-07-25 06:59:55.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra.tilay@hpe.com","date":"2022-08-22 11:44:53.000000000","tz":0},"subject":"HPE 3PAR: Fix umanaged volumes \u0026 snapshots missing","message":"HPE 3PAR: Fix umanaged volumes \u0026 snapshots missing\n\nThe umanaged volumes \u0026 snapshots are missing when below commands\nare executed:\ncinder manageable-list\ncinder snapshot-manageable-list\n\nTo address this, added two functions:\nget_manageable_volumes()\nget_manageable_snapshots()\n\nCloses-Bug: #1819903\nChange-Id: Ie178e48958a69ef2b6c329e132de2c772d5c1a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d483ad49e9b438b01e1639664b1d77a74edcf45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d483ad49e9b438b01e1639664b1d77a74edcf45"}]},"branch":"refs/heads/master"},"c112542cf5471641a47a94c89b9b80dfa14d701d":{"kind":"REWORK","_number":7,"created":"2022-09-06 12:08:35.000000000","uploader":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"ref":"refs/changes/36/850836/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/36/850836/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/36/850836/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/36/850836/7"}}},"commit":{"parents":[{"commit":"a2deecef891a16703e5410eb7805442961307090","subject":"Merge \"mypy: service.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a2deecef891a16703e5410eb7805442961307090"}]}],"author":{"name":"raghavendrat","email":"raghavendra.tilay@hpe.com","date":"2022-07-25 06:59:55.000000000","tz":0},"committer":{"name":"raghavendrat","email":"raghavendra.tilay@hpe.com","date":"2022-09-06 12:03:07.000000000","tz":0},"subject":"HPE 3PAR: Fix umanaged volumes \u0026 snapshots missing","message":"HPE 3PAR: Fix umanaged volumes \u0026 snapshots missing\n\nThe umanaged volumes \u0026 snapshots are missing when below commands\nare executed:\ncinder manageable-list\ncinder snapshot-manageable-list\n\nTo address this, added two functions:\nget_manageable_volumes()\nget_manageable_snapshots()\n\nCloses-Bug: #1819903\nChange-Id: Ie178e48958a69ef2b6c329e132de2c772d5c1a40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c112542cf5471641a47a94c89b9b80dfa14d701d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c112542cf5471641a47a94c89b9b80dfa14d701d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
