)]}'
{"id":"openstack%2Fcinder~850840","triplet_id":"openstack%2Fcinder~master~Ifbb1d7d5271ef62cc12f4de8d916661bdba67640","project":"openstack/cinder","branch":"master","topic":"fujitsu-migrate","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2023-06-07 10:06:36.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"30615":{"account":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"last_update":"2023-06-07 10:06:36.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"33609":{"account":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"last_update":"2023-09-15 01:16:10.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ifbb1d7d5271ef62cc12f4de8d916661bdba67640","subject":"Fujitsu Driver: Add update migrated volume support","status":"ABANDONED","created":"2022-07-25 08:00:16.000000000","updated":"2023-09-15 01:16:10.000000000","total_comment_count":35,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"11969252042761dbab99294fb73036ddacd2847c","_number":850840,"virtual_id_number":850840,"owner":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-06-07 11:41:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"all":[{"value":1,"date":"2023-06-07 10:06:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":1,"date":"2023-07-05 03:52:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-25 08:04:35.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-07-25 08:15:32.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-07-25 08:25:27.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 08:27:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 09:06:35.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-07-25 09:13:25.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-07-25 09:27:37.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-07-25 09:57:25.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 10:38:47.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-07-25 10:42:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-25 10:56:29.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 11:32:24.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 11:40:21.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 20:30:20.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-07-27 12:33:34.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-07-27 13:13:09.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-07-27 13:14:25.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-07-27 15:00:29.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-27 18:54:35.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-28 02:16:44.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-07-29 02:05:36.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-08-04 11:10:46.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-21 05:50:25.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-22 07:07:56.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2023-03-22 03:20:53.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2023-03-24 03:36:20.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-03-24 03:56:24.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-24 09:51:09.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-03-24 14:31:43.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-03-24 15:28:40.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2023-03-28 00:27:14.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-05-30 10:37:28.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2023-05-31 22:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-06-07 10:06:36.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2023-06-07 18:01:39.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"}],"messages":[{"id":"bba7c9e8ce1b0f4d2d9a7b7ac7ce20c700e0352a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-07-25 08:00:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7b1f3fb7c132719a660750c7d6211773155392e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-25 08:04:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"4916432d0ea0c52131d76d776f83c9ba9b508dce","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-25 08:15:32.000000000","message":"Patch Set 1:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/230","accounts_in_message":[],"_revision_number":1},{"id":"40732a2f47a0eed10fd82c58c1097b8ad5e17b15","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-25 08:25:27.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"ca7c97d3841b35fe12b0a816c96178f44feea619","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 08:27:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c3d00f4493a464e90a744ebb62072e0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5208d921f2d24bb0a2400a7bf7cdcc59 : FAILURE in 25m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9d0749dea254f34b860851bc3c36b87 : FAILURE in 19m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29043dd6b0d1b6ec0fc8e092e269231bbbd3c373","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-25 09:06:35.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850840/1/iscsi/ : SUCCESS in 1h 05m 24s\n- cinder-smb http://cloudbase-ci.com/cinder/850840/1/smb/ : SUCCESS in 51m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d914a61ae034fdf7bcfb8fb21795a2b55f41ccab","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-25 09:13:25.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1658736025861 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"c40519dd1d515e3cd1dccc938799515d27d643cc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-25 09:27:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/40/850840/1/check/iscsi-driver-alletra-6k/b4bcefd/b4bcefd.html : FAILURE in 1h 02m 54s","accounts_in_message":[],"_revision_number":1},{"id":"43ad933e116bf1021e94d163c0d1e08afcc6bc67","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-25 09:57:25.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/c1c8d11d35064ea8bfbad56afd8edaae : POST_FAILURE in 1h 43m 09s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/816089ef129e4fb880010b8698577d4e : POST_FAILURE in 1h 30m 53s","accounts_in_message":[],"_revision_number":1},{"id":"dba2986b90fbf2f7998692360a93e9dc750d2069","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-25 10:38:47.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/40/850840/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/4171026/ : SUCCESS in 1h 22m 48s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/40/850840/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/ea2f3f1/ : SUCCESS in 1h 39m 29s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/40/850840/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/642f0dd/ : FAILURE in 30m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0ec72a864e06377ddb5952908bfadf43ab3f08d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 10:42:47.000000000","message":"Patch Set 1: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/65af98010b4840318435acc11befcf6d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6b1e5e6d31b4699b791a48602b61e25 : FAILURE in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d41fc445c9d4f5086cf77990fb127c4 : FAILURE in 12m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80e05e7a1a80426f9bae301d0cb87d5f : FAILURE in 12m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ccd8cf29fd94111ba6299936237cef3 : FAILURE in 12m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc2171795ee44f5598d9a0db34d494c1 : SUCCESS in 12m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/668a15a6554c43c0b30ad03eff186070 : SUCCESS in 1h 05m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cd7ada9fdc674db7be57375655fb238f : SUCCESS in 1h 28m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e64dd4557df47249e6e0ce52cca2e15 : SUCCESS in 1h 01m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/56e7ebab768c4cbbb576eb436976747f : FAILURE in 12m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6ab078e173ba4079ab04690892b8be20 : FAILURE in 5m 33s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d44be3340af34fc799cc3de732b7b9c4 : FAILURE in 6m 36s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cca7c4de49504fb39b602d6d834822e6 : SUCCESS in 6m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/25401c9281cb40a68df18b1570acd384 : SUCCESS in 6m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b06dc87c2ce3462e8c5814445a66065c : SUCCESS in 46m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3d0805b0a17746e7a90314b5c5abac02 : FAILURE in 6m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8a9a6b91051d4e3ca83b44cae6cd44b5 : SUCCESS in 1h 31m 01s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/442734828a284e7f87b31f67d2743b16 : FAILURE in 1h 16m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a1f14edb48e84c5fbe86ffa86f0056cd : SUCCESS in 1h 50m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c8e645d55b09484ea64ca49e6371abb0 : SUCCESS in 1h 16m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5157d5f5efb04f9888ed789a01b961e0 : SUCCESS in 1h 26m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1e56925992654d87a5beeda574bb7ec9 : SUCCESS in 35m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/20e70ac62f694cf09961bcd9aa1e3b7d : SUCCESS in 1h 37m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1378551657f148108a19ce08b5e35046 : SUCCESS in 2h 35m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2e1c7f04c684fea8460edef0a89a59a : SUCCESS in 43m 11s","accounts_in_message":[],"_revision_number":1},{"id":"302d2996e69195547c00a61dd80ec1bb4b6a52a9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-25 10:56:29.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850840 : SUCCESS in 1h 35m 01s","accounts_in_message":[],"_revision_number":1},{"id":"3358f22d10e474f55617fbdac19d34667df8962f","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-25 11:32:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/1/check/fujitsu-eternus-dx-iscsi/d1f93cc : FAILURE in 1h 23m 04s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/1/check/fujitsu-eternus-dx-fc/c7f4539 : FAILURE in 1h 41m 22s","accounts_in_message":[],"_revision_number":1},{"id":"28d5d15fe3f9cd33ca9ce8aee6ed7f03e081d911","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-25 11:40:21.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-40-850840-1/ns5_nfs : SUCCESS in 3h 22m 48s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-40-850840-1/ns5_iscsi : SUCCESS in 3h 24m 10s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"effeb7485271f8ee562f6b67bae3b4202ee5e8d9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-25 20:30:20.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/40/850840/1/check/DellEMC_Unity_FC/eb2e3af/DellEMC_Unity_FC/1245 : FAILURE in 19m 02s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/40/850840/1/check/DellEMC_Unity_iSCSI/36fc213/DellEMC_Unity_iSCSI/1151 : SUCCESS in 2h 00m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"1e1c60e80684d76b4861ad8e63ded804d2c0a624","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-07-26 01:46:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"98d76cabc021707c0249eada8cfcc99d68230078","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-26 02:27:19.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1658799968427 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"0ce4ba42469c79866f2afb13b1e50546cd5ccf4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 02:49:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c220e874efae4702b72f23e5fca34361\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0883aa1cce3f4400aaa983fe7fad5897 : FAILURE in 13m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70884ba9c8124890b43fc2a2e776ef68 : TIMED_OUT in 1h 01m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ac60d4ff342e0ad61836a7192f3ee15897d200ca","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-26 02:49:56.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/40/850840/2/check/iscsi-driver-alletra-6k/1807805/1807805.html : FAILURE in 1h 02m 42s","accounts_in_message":[],"_revision_number":2},{"id":"a336293cd8cd5a8383e027435a0442943f6186ce","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-26 02:58:01.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/2/check/fujitsu-eternus-dx-iscsi/8abd50b : ABORTED in 1h 11m 24s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/2/check/fujitsu-eternus-dx-fc/8d71af5 : ABORTED in 1h 11m 20s","accounts_in_message":[],"_revision_number":2},{"id":"e1ec230ea4ecd2f63a5951af3c6cbad315e2605f","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-07-26 03:08:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"20b0af6e83bced10161f474182f34943dedc0109","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-26 03:28:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850840 : FAILURE in 1h 40m 01s","accounts_in_message":[],"_revision_number":2},{"id":"dd70d2b025996edceaf76ebf0c62f52f698637de","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-26 03:28:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850840/2/iscsi/ : SUCCESS in 1h 42m 13s\n- cinder-smb http://cloudbase-ci.com/cinder/850840/2/smb/ : SUCCESS in 1h 11m 38s","accounts_in_message":[],"_revision_number":2},{"id":"7a1d0560aa3603d6c14593c0e6902f497ac68be4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-26 03:35:05.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/770f2a4831d344bcbcd5ba2339449a1f : POST_FAILURE in 1h 19m 55s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/4565c46295124b3c81558b636049ab4a : POST_FAILURE in 1h 45m 47s","accounts_in_message":[],"_revision_number":2},{"id":"1c697e4af1446c4d54af5e8bed467b660077192a","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-26 03:38:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/2/check/fujitsu-eternus-dx-iscsi/78a1511 : ABORTED in 28m 39s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/2/check/fujitsu-eternus-dx-fc/aea6aa3 : FAILURE in 24m 47s","accounts_in_message":[],"_revision_number":2},{"id":"0b6ce94453df86eb14f0dbc6e3ec713b663ace05","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-07-26 03:39:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ed41eb67973b9ffa128d7ab92cb6d521c647ae62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 04:10:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/556529fc01af4364a75f07b2f1be9343\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d5d2ef64b2344b6acac28c8619fa587 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a23faff6490647de95119fc6781b1c9b : FAILURE in 12m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2676576f40014616a868271f0af9a91f : FAILURE in 11m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf71f69d8ea84b019986cd0a6b63122e : FAILURE in 10m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9927573bb1443cab82ffd7a308edfa3 : SUCCESS in 10m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/76ae126b0695465587c16fe318b2bb45 : SUCCESS in 1h 07m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9ddaec21e1284329b496d6cf2811cda2 : SUCCESS in 1h 29m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4107ea473ccf420fa3c0db2060e93a62 : SUCCESS in 51m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2ac714ed59d3470cb479ce7c0acb34c4 : FAILURE in 14m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/249df33b5f434d318299342890b0b428 : FAILURE in 5m 18s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/134bbd60c4b349cfb4d6eeebe11ba8e4 : FAILURE in 6m 19s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eac120a9065b48abb8327912b80d4012 : SUCCESS in 6m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b4e5d2a8ef5542f08558969720956004 : SUCCESS in 5m 38s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dac90bb2e6c845cca8b4ee199eac1519 : SUCCESS in 45m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/29b3beca36de4581b56bf34ab9abb59f : FAILURE in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fc06f07eacad44c58a7cae3ecb4eb9e1 : SUCCESS in 1h 33m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/921744196ecd465bb910d11de6bf1bfb : FAILURE in 1h 40m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f4660a07d9f84afeaa165aac1273cbee : SUCCESS in 1h 43m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ab253ab060994b178c7dde3d0c236624 : SUCCESS in 1h 20m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b043c87119584b00abdfc256ac36a408 : SUCCESS in 1h 44m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b898c8ee89874bdcafb5986d6d2943f3 : SUCCESS in 46m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/250eccb3292941e0b5586b537f7a0ecf : SUCCESS in 1h 30m 50s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9321b3a0fd614b5eb7264a297fce3ef6 : SUCCESS in 2h 17m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13b707aaa28c48b6b1f5b07b51f6429b : SUCCESS in 1h 07m 27s","accounts_in_message":[],"_revision_number":2},{"id":"c552682f79f8732567688a636647685f2eac9880","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-26 04:29:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/40/850840/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/c198b8c/ : SUCCESS in 1h 36m 56s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/40/850840/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/22df879/ : SUCCESS in 1h 24m 22s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/40/850840/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/1890a9d/ : FAILURE in 31m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"59ce4b076be670234e521345766500b7be894ac6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-26 05:10:57.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-40-850840-2/ns5_nfs : SUCCESS in 3h 21m 50s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-40-850840-2/ns5_iscsi : SUCCESS in 3h 24m 26s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"d841e1c3b1b0a060c8653c85d54c6935c3c2862e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-26 05:24:28.000000000","message":"Patch Set 2:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":2},{"id":"a4465f0b4f1c72df510fb98851d1c0e3458c7ce1","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-26 06:20:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/2/check/fujitsu-eternus-dx-iscsi/64f9a50 : ABORTED in 2h 39m 40s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/2/check/fujitsu-eternus-dx-fc/c00d5de : ABORTED in 2h 39m 38s","accounts_in_message":[],"_revision_number":2},{"id":"13d864e4b62a2d875856eb164eb58b941a6e542f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-07-26 06:44:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e0b181d3066b4c6ea7b218f0c1c40da90d4abb9f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-26 06:51:40.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"e28652ed7bd009a43f1833e3b37717e630a7bc16","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-26 07:01:09.000000000","message":"Patch Set 3:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":3},{"id":"8063b87e72132e08ae57ed42fcf1a30ec0331b7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 07:06:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b490a06b6e4143ad9bd3642c0a3e7c5e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba55185f90c54f81aa0f0b97ee574a91 : SUCCESS in 19m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8d3e09c62b274c978cbad6e691dd9ba5 : SUCCESS in 19m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5392bd838645182ca235ab3d0b58e6f33745d087","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-26 07:22:49.000000000","message":"Patch Set 3:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1658817893933 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"39e71da2c7a22427a4c7d84b9efb9284043d6b71","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-26 07:48:52.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/40/850840/3/check/iscsi-driver-alletra-6k/4950d8b/4950d8b.html : FAILURE in 1h 03m 01s","accounts_in_message":[],"_revision_number":3},{"id":"fd2b3f260d47abca217beaa0067213325b09c9bc","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-26 08:01:00.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850840/3/iscsi/ : SUCCESS in 1h 15m 43s\n- cinder-smb http://cloudbase-ci.com/cinder/850840/3/smb/ : SUCCESS in 53m 49s","accounts_in_message":[],"_revision_number":3},{"id":"ff56a0f263fac6d16124950f48bf824de94e90bb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-26 08:17:17.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850840 : SUCCESS in 1h 31m 01s","accounts_in_message":[],"_revision_number":3},{"id":"a590b1abcf8527acdd23ee98c6c638fb17341a31","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-26 08:27:29.000000000","message":"Patch Set 3:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/3/check/fujitsu-eternus-dx-iscsi/2cbc2b5 : SUCCESS in 1h 42m 08s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/3/check/fujitsu-eternus-dx-fc/2d2222c : SUCCESS in 1h 40m 27s","accounts_in_message":[],"_revision_number":3},{"id":"90a3ef2d81380163b8db7bd96fe89e796470bbf7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-26 08:42:27.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/40/850840/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/e9be44a/ : SUCCESS in 1h 56m 40s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/40/850840/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6bb1771/ : SUCCESS in 1h 38m 03s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/40/850840/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/840a188/ : FAILURE in 28m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"05442ee15f6c20b12d01e0ab193190447a7fd81a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 08:50:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4594eaf9a87b4521bcb312c043b9abdb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aea7708f3fc146caa8a5c3ebc4af3f0d : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4e5b69835d3484987361ce3435d913b : SUCCESS in 11m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1b09f35e2d34cd7867b755ec6a94493 : SUCCESS in 15m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9f7d83430ed409f9d8e9398ec636a6c : SUCCESS in 11m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/978cad7c06584be684d28bef021c4651 : SUCCESS in 11m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/487419153ca94d3eb007c102cbd0ae06 : SUCCESS in 57m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cb7eb9135386449da675869a0d959ba7 : SUCCESS in 1h 21m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/287cda0934f949b1b41a4dcaed5cfebe : SUCCESS in 50m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c611e759b3f2449fb6bd2ca6bc71dd4f : SUCCESS in 14m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cdc689e865694b248d1782559c115ed2 : FAILURE in 5m 46s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/02cacebfa3264397a81bbda25c7b01ed : FAILURE in 7m 19s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6d1cc14becaf4fc49d2052897c85cd18 : SUCCESS in 7m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f3c23c1ec2904cb5a13ae75ba5e299fb : SUCCESS in 7m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5a55e9b6ee0841629f3bff065e815f55 : SUCCESS in 34m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/64c95e60e75e4a86be9899d998ec5cf2 : FAILURE in 6m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fd6f628b376f42fb9b38ac3ff27c606b : FAILURE in 1h 28m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cf7f54ba8c2d463aa586898dd0f2555b : FAILURE in 1h 22m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d830019beb764b9bbdd128281b1ba3fc : SUCCESS in 1h 38m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1c27cc95bb974baea37fdba6ef9c2390 : SUCCESS in 1h 11m 48s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ca984798728b4584b0471060c9737a4e : SUCCESS in 1h 27m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7bcdfe2f17f34bf493637c22766a89df : SUCCESS in 46m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a30116f7d410429280d71e40528d54af : SUCCESS in 1h 31m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8747a62ce44440a4b9e4213c6dd42cca : SUCCESS in 2h 03m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0505de80e5ab41199681d7ced7e0533e : SUCCESS in 53m 19s","accounts_in_message":[],"_revision_number":3},{"id":"bbb7fa224ccb2ba74f3943dfb4aa264faf637ba6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-26 08:51:38.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/36ec2c4e74924619a35feb8901e66659 : POST_FAILURE in 2h 03m 21s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/2e9578d0eb8c4d53bdda2900ea376c03 : POST_FAILURE in 1h 41m 13s","accounts_in_message":[],"_revision_number":3},{"id":"bb332fbb69167c79498864895c2210e366470c13","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-26 10:08:35.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-40-850840-3/ns5_nfs : SUCCESS in 3h 22m 12s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-40-850840-3/ns5_iscsi : SUCCESS in 3h 23m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"4dde786acbb38c03558b6d57f15f60472df9fa23","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-26 10:11:08.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/3/check/fujitsu-eternus-dx-iscsi/c74f136 : NOT_REGISTERED\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/3/check/fujitsu-eternus-dx-fc/7d2636e : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"2d1db528a90c6ec885c1276d677030b26ce61a07","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-26 10:16:05.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/3/check/fujitsu-eternus-dx-iscsi/1373888 : NOT_REGISTERED\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/3/check/fujitsu-eternus-dx-fc/a8cea0f : ABORTED in 59s","accounts_in_message":[],"_revision_number":3},{"id":"4c18247d3d78259babd82ab4850cf0e5da34a34e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-27 02:22:46.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/40/850840/3/check/DellEMC_Unity_FC/2b08b2e/DellEMC_Unity_FC/1257 : FAILURE in 20m 33s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/40/850840/3/check/DellEMC_Unity_iSCSI/f4ac1bb/DellEMC_Unity_iSCSI/1158 : SUCCESS in 1h 54m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"735798e8ee1d9255754a14f16298a774c59b3cd6","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-07-27 10:03:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c961587e330232fc7cd1cbddcef9176736ddbcaf","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-07-27 10:04:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fc1512d203eb3d208dfcffdca6961ed2b054b819","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-27 10:08:34.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"aeecd4275f3c1c0676afa55b4048421f70d5aac5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 10:41:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/431073c579f241f3884104b3ff06b5d1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b8adc1255504df8a07c42c8e01a6b1e : SUCCESS in 27m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a3556d11fd124ace9b357c3dbe3f29d1 : SUCCESS in 25m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d6a0eb8378edf119ffb01a70948c7b55c5f3de35","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-27 11:11:07.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/850840/3/iscsi/ : SUCCESS in 1h 06m 39s\n- cinder-smb http://cloudbase-ci.com/cinder/850840/3/smb/ : SUCCESS in 1h 07m 09s","accounts_in_message":[],"_revision_number":3},{"id":"6b5ab4e641e4abeedc9198bdd84042015e5e4da7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-27 11:32:30.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/40/850840/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/8469d4f/ : SUCCESS in 1h 28m 11s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/40/850840/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/b6ff392/ : SUCCESS in 1h 25m 18s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/40/850840/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/b7ec8db/ : FAILURE in 31m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1956b8d95a817dd979f2e7bfdcfbe86df343b655","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-27 11:46:48.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/3/check/fujitsu-eternus-dx-iscsi/97e3124 : FAILURE in 1h 41m 13s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/3/check/fujitsu-eternus-dx-fc/f8b64b1 : FAILURE in 1h 31m 13s","accounts_in_message":[],"_revision_number":3},{"id":"e2255c52b710a7b631905b4877ae4bdcf1c61b32","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-27 12:24:20.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/dbf596f290764c6995b7b2c94863fa54 : POST_FAILURE in 1h 44m 05s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/45c01e91a592425a89bfa079aea46297 : POST_FAILURE in 1h 50m 06s","accounts_in_message":[],"_revision_number":3},{"id":"d265e0a6fc3b796c179975852cda41e56c99053d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 12:25:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce48a6055a464508bcf3c37515dcb1f2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9a66ba9612c45ca88494ab14b1fca10 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50ab893825e9445fb0f8ca48eff8fcc6 : SUCCESS in 11m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82e67cef9fdf44089f702cb0bc4b0fa5 : SUCCESS in 11m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc03b7aa77ce4a36909a6596ea808132 : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b454bff81f5472692e45f93d0bfd190 : SUCCESS in 12m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/867fdfdef7fc4ae5a6c99b66aa142ab8 : SUCCESS in 1h 05m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4976db911ddd413e945d3d2fbf0b5fd1 : SUCCESS in 1h 20m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0197dc7e3e54427d99b3b6192e561939 : SUCCESS in 52m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6cb9c5d3af6c408185ee5338d55f9ce9 : SUCCESS in 15m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/64c82d6042274ff199703d3466cf9675 : FAILURE in 5m 25s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/da2b02624eb5497581366249c27f9304 : FAILURE in 6m 53s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ac3f834073a64e62a14f5edf59f78bea : SUCCESS in 6m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b7cd62794b17453fb224a42a26bacd1c : SUCCESS in 6m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0a41d3736cdf489e9c141acaa64e1a8b : SUCCESS in 34m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1598bfea56814712b148afe3fd8570ae : FAILURE in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9857453a89b34d488dbea3f6c5a88cf5 : SUCCESS in 1h 24m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3a1666369d6c4854bd9f85ed70d8a6ea : FAILURE in 1h 39m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6a9e1058af3c41258326703e1500191e : SUCCESS in 1h 46m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0a86b8f5b5ac42ce944f66652753110f : SUCCESS in 1h 16m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/625c3a3760f4459b814a7de853296c5c : SUCCESS in 1h 41m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d25510431cd04a14a3e1456a1b75620c : SUCCESS in 45m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5ff3863cda4c47908600f5bc58893957 : SUCCESS in 1h 33m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5a42401bd6ff492d872e26f21934cd8e : SUCCESS in 2h 16m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64bee5310d56416cb887acdafe787fe4 : SUCCESS in 57m 25s","accounts_in_message":[],"_revision_number":3},{"id":"90bbd3c83f51068ad647c76f7b6832d4f856351d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-07-27 12:33:34.000000000","message":"Patch Set 3:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/27_07_2022_12_25_36\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"3495f2c650bb1fb66f108139e13e19afba3df397","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-27 13:04:39.000000000","message":"Patch Set 3:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/286","accounts_in_message":[],"_revision_number":3},{"id":"5deb5a826c630d8755bf78978af268c1a046f6f7","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-07-27 13:13:09.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/40/850840/3/check/DellEMC_PowerStore_NFS/78539bc/DellEMC_PowerStore_NFS/187 : ABORTED in 4m 19s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"e666da48a541d551b596853dea49aaba3c83f6ce","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-27 13:14:25.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/850840/3/upstream-check/cinder-tempest-plugin-ontap-fcp/e4a3581/ : RETRY_LIMIT in 9m 53s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/850840/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/c28b01f/ : FAILURE in 47m 47s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/850840/3/upstream-check/cinder-tempest-plugin-ontap-nfs/60e3a52/ : SUCCESS in 42m 53s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"06767f0b9cdab5a37f185467d7caf9bf6d5b3865","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-07-27 15:00:29.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/40/850840/3/check/DellEMC_PowerFlex/9606ea7/DellEMC_PowerFlex/1086 : SUCCESS in 1h 20m 56s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"9500a46b2570f36d56cafb858baf3edd2b04e9dc","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-07-27 18:54:35.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/40/850840/3/check/DellEMC_PowerMAX_FC/6463043/DellEMC_PowerMAX_FC/759 : FAILURE in 44m 31s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/40/850840/3/check/DellEMC_PowerMAX_iSCSI/6675714/DellEMC_PowerMAX_iSCSI/742 : FAILURE in 49m 39s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"16dd5019b723436a33b04d2e164877c9233e2a84","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-07-28 02:16:44.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/40/850840/3/check/DellEMC_SC_FC/cf6c1b3/DellEMC_SC_FC/818 : SUCCESS in 51m 56s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/40/850840/3/check/DellEMC_SC_iSCSI/6c31d61/DellEMC_SC_iSCSI/838 : SUCCESS in 1h 58m 39s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"22ec0246d29bb50c8afea894cf46fe82b23763f0","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-07-28 02:50:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"35a3cb6edf9be451b60171449528f66049b32b8d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-28 04:47:30.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/3/check/fujitsu-eternus-dx-iscsi/aaa8ef7 : FAILURE in 1h 45m 01s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/3/check/fujitsu-eternus-dx-fc/3a2e5d0 : FAILURE in 1h 55m 59s","accounts_in_message":[],"_revision_number":3},{"id":"0210733dd611843d8c11b07d95bfc06e5d50bc13","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-07-28 05:25:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dc91a3781ce53d9edde467bd31b2aa5226236253","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-28 07:02:02.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/3/check/fujitsu-eternus-dx-iscsi/56560d9 : FAILURE in 1h 35m 38s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/3/check/fujitsu-eternus-dx-fc/04e8723 : FAILURE in 1h 34m 58s","accounts_in_message":[],"_revision_number":3},{"id":"ca33bba7849a5157a9324498d22c37db50d6d91a","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-07-28 08:41:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"372d9b45e2ce74ea597ec16476161b3ac7975bfa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-28 09:48:04.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/40/850840/3/check/DellEMC_Unity_FC/f32b59e/DellEMC_Unity_FC/1279 : FAILURE in 22m 09s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/40/850840/3/check/DellEMC_Unity_iSCSI/9adac27/DellEMC_Unity_iSCSI/1172 : SUCCESS in 2h 03m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"04f38fd8c887a215cd0541940412630469792af8","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-29 02:05:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/40/850840/3/20220729084736/NEC-V-FC : SUCCESS in 2h 05m 34s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/40/850840/3/20220729084736/NEC-V-ISCSI : SUCCESS in 2h 17m 57s","accounts_in_message":[],"_revision_number":3},{"id":"b06c32ded4838df81b36838c4779582d5cbffba1","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-01 11:28:56.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/3/check/fujitsu-eternus-dx-iscsi/5c484dd : FAILURE in 1h 28m 08s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/3/check/fujitsu-eternus-dx-fc/be80d88 : FAILURE in 1h 34m 35s","accounts_in_message":[],"_revision_number":3},{"id":"9bb0ee0d11c0817e3e0099ccc35b2bf5029ef2ad","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-01 14:27:01.000000000","message":"Patch Set 3:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/3/check/fujitsu-eternus-dx-iscsi/5cf9b09 : SUCCESS in 1h 43m 56s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/3/check/fujitsu-eternus-dx-fc/d3b05a0 : SUCCESS in 1h 51m 29s","accounts_in_message":[],"_revision_number":3},{"id":"aef3b3dd34d154a706403901fcbd14fb459e6eab","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-02 03:28:19.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/3/check/fujitsu-eternus-dx-iscsi/76af3e7 : SUCCESS in 1h 38m 21s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/3/check/fujitsu-eternus-dx-fc/6899a7b : FAILURE in 1h 37m 38s","accounts_in_message":[],"_revision_number":3},{"id":"b6ff3324bb8bebbb3f2bd3de1589feb59f2259b0","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-08-02 03:36:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9000621914b1ce7ac92a71c01c253ff73c394313","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-02 05:20:43.000000000","message":"Patch Set 3:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/3/check/fujitsu-eternus-dx-iscsi/9bd5c2f : SUCCESS in 1h 35m 09s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/3/check/fujitsu-eternus-dx-fc/1e759fa : SUCCESS in 1h 43m 32s","accounts_in_message":[],"_revision_number":3},{"id":"352322c159f8bc9997460eedeb8759d664005959","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-08-04 11:10:46.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/40/850840/3/20220804164124/NEC-ISCSI : SUCCESS in 3h 28m 48s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/40/850840/3/20220804164124/NEC-FC : SUCCESS in 3h 29m 18s","accounts_in_message":[],"_revision_number":3},{"id":"6f225696df28f1ce5b0a86a006334e4b6b84cad6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-11-21 00:57:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4a9ec4db98246921cc219e8dd7102af4f5a01b34","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-21 00:57:12.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1752 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"c0478172b9c39bd5cc6c8de676c63390498f54b8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-21 00:58:54.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"52c1168ab92eb83229f875598fa1638b7fd784ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 01:23:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b63e56867a2b4c88b4e72c19bcef66bf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/460399fea4f643d8a7ffd42dcb75701c : FAILURE in 20m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5114be0dc6564511b331657c74b2ae80 : FAILURE in 20m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8dff96b460a496adcc77d85216489b322f13941e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-21 02:27:27.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/850840 : SUCCESS in 1h 28m 42s","accounts_in_message":[],"_revision_number":4},{"id":"6869696a5d9d087d34d79903af3fdb3b1de30856","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-21 02:37:10.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850840 : FAILURE in 1h 36m 22s","accounts_in_message":[],"_revision_number":4},{"id":"bfe43e47e3737e47f7792b20eae0e0a45d9cedd1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-21 02:41:55.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/40/850840/4/check/iscsi-driver-alletra-6k/1133196/1133196.html : FAILURE in 1h 43m 14s","accounts_in_message":[],"_revision_number":4},{"id":"a033a5f3ea33e59e7ee378b74b74666a410184d4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-21 03:04:29.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/40/850840/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/9156b56/ : SUCCESS in 1h 26m 34s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/40/850840/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/cdaeb2b/ : SUCCESS in 1h 37m 21s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/40/850840/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/cb4c7b3/ : FAILURE in 2h 06m 21s","accounts_in_message":[],"_revision_number":4},{"id":"e6fa74109b45970bec837bdd95406151f2edc679","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 03:13:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1cbddba3aa243fab1e655fedd7b72ae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91e9a67d6e1547e68ef1cb772289db45 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22c8200083344b2db351847b27d1a9af : FAILURE in 9m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a8cb4ed8b1e414382849053161e0af7 : FAILURE in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ca8362737b34f84ae57af500c4566e8 : SUCCESS in 12m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/8e7ee1c50da84791aeed5ddad70bf661 : SUCCESS in 1h 00m 14s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fec4d9f5dff1448facd3b8a4835861ad : SUCCESS in 1h 08m 18s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ce92464505a4444487da223ba9aee016 : SUCCESS in 1h 34m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/568446354c0b4777a1d5314082d321fa : SUCCESS in 57m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c9f79e3862f54ebf955cd5ce9ab207d5 : FAILURE in 12m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d49101e5658c4a708a48c7ebde72764e : SUCCESS in 5m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/78718ed17bd1466ea8da9f3ec36f0ffe : FAILURE in 6m 15s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9fefff5fffbf44feb5d846a4f021f7b2 : SUCCESS in 5m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/756a3cde09ed40f293b63d0ee689f913 : SUCCESS in 5m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/96bf24abe27743bab905a3af5bba3443 : SUCCESS in 51m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0ab2975c2b634ab4a5990775417a0164 : FAILURE in 6m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8e54e01c88f948a1ace9c18850733e63 : SUCCESS in 1h 20m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4085507d05324e8583ada862f2d64936 : FAILURE in 1h 12m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bc403ad0361e4eedab2608e8a64d6485 : SUCCESS in 1h 41m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/df07d48dd8054fe29cdb06558020c605 : SUCCESS in 1h 10m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/94e9d74c0b92433e9f3105cc6de9cca0 : SUCCESS in 1h 33m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fe512b3d9e044d91a481585707707e10 : SUCCESS in 42m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f02abe5b58fa46ae8f4830be024f1b96 : SUCCESS in 1h 32m 19s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1fe95e64afc740908fe1b0a43cb7abb7 : SUCCESS in 2h 11m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49f5b57f9873491081dce64d99be7dcf : SUCCESS in 51m 10s","accounts_in_message":[],"_revision_number":4},{"id":"8cb13235f36946924550a4a216f8f3e3279ee4b1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-21 04:32:31.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-40-850840-4/ns5_nfs : SUCCESS in 3h 29m 0s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-40-850840-4/ns5_iscsi : SUCCESS in 3h 34m 57s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"c3cde0de1f794e49d0a32ba6ed8c995fe66ebf61","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-11-21 04:32:59.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/40/850840/4/20221121095705/NEC-ISCSI : SUCCESS in 3h 34m 59s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/40/850840/4/20221121095705/NEC-FC : SUCCESS in 3h 35m 51s","accounts_in_message":[],"_revision_number":4},{"id":"e5b787f7c7aef789c26e2d46912e2b24ce6e6e6f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-11-21 05:50:25.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/40/850840/4/ : FAILURE in 1h 29m 18s","accounts_in_message":[],"_revision_number":4},{"id":"0f2d560ad0a73af04f23eb2557175e8142aba6a5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-11-21 06:46:36.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/40/850840/4/20221121095707/NEC-V-ISCSI : SUCCESS in 2h 51m 37s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/40/850840/4/20221121095707/NEC-V-FC : SUCCESS in 5h 49m 25s","accounts_in_message":[],"_revision_number":4},{"id":"9b71147e7a126bcc4a5deb4ef4c0030ff37d17e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-11-22 03:59:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"96a134d121135c857c9650f81d92f717d92ada03","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-22 03:59:20.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1763 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"9b52c7446da0daf063691fa8cc54bf77eebb6c34","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-11-22 04:07:21.000000000","message":"Patch Set 5:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/842","accounts_in_message":[],"_revision_number":5},{"id":"602a89599300ef0c92ddc00d35e266366c25e1e3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-22 04:17:48.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"a4c7ecc6cb5f07505d6b4fee279cf077d6a5b930","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 04:26:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00ae247c76dd4dfaa16c49a00eeaddcf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0dfec7d793ce4935a1c5d9bfc747ec0d : SUCCESS in 21m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cca2b6b6276e4ff1aff692f0a9310fb7 : SUCCESS in 21m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2ba2905d5f1a6ed807184c79fcfd1bc13ddd8993","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-22 05:04:55.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/40/850840/5/check/iscsi-driver-alletra-6k/622e435/622e435.html : FAILURE in 1h 04m 41s","accounts_in_message":[],"_revision_number":5},{"id":"e508935dfa47c9d7e7bbcdb92dc7c46275692a5c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-22 05:19:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"92e310e793176b58374f9b08fb52da267f29ef38","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-22 05:22:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5ce17fcf51e7b580aee478ed86144ae612b1849c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-22 05:32:09.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/850840 : SUCCESS in 1h 30m 30s","accounts_in_message":[],"_revision_number":5},{"id":"a29e7271dcd310360c4f23d5bd4d2a583d910721","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-22 05:36:35.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850840 : SUCCESS in 1h 34m 26s","accounts_in_message":[],"_revision_number":5},{"id":"cd4c63728a79a121a050aacad4ef4487dfaaa4a0","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-11-22 05:51:17.000000000","message":"Patch Set 5:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/5/check/fujitsu-eternus-dx-iscsi/ce51e4d : SUCCESS in 1h 50m 42s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/5/check/fujitsu-eternus-dx-fc/e408a3e : SUCCESS in 1h 51m 27s","accounts_in_message":[],"_revision_number":5},{"id":"717c39d936c6b70530bf0babd3b2a3d1069dc133","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-22 05:57:57.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/40/850840/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/c03f3de/ : FAILURE in 1h 58m 01s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/40/850840/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/de3fd9d/ : SUCCESS in 1h 38m 26s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/40/850840/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/04db072/ : SUCCESS in 1h 43m 31s","accounts_in_message":[],"_revision_number":5},{"id":"65beef84918b35b19f39d19e6c112af3b4cecfe0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-22 06:02:03.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/40/850840/5/check/iscsi-driver-nimble/328fb09/328fb09.html : FAILURE in 1h 06m 27s","accounts_in_message":[],"_revision_number":5},{"id":"b891f87ff394c0b8138163d8c2e4afdfa9d33749","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-22 06:12:39.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 13m 39s","accounts_in_message":[],"_revision_number":5},{"id":"406af11c68a3975db803e215570dc2c8be132e3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 06:17:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d43604ebb45f4739bb1867a6e75b8af2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11cb67fcc9f542e7bcb8c6e457ae0653 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96df5b19e6f24eaea8185a920fd765ed : SUCCESS in 10m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fda5bfb8ea7347e5a0dd77f6b1634848 : SUCCESS in 9m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d68148f675e946568f47f5affabed0d8 : SUCCESS in 11m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/7cef1555bc4d43998facfe3622e4147a : SUCCESS in 1h 10m 25s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d3d6b6dca8774fef8260408c32afb0c1 : SUCCESS in 1h 03m 27s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6e737b82ee834f028cb35a36a3d59de4 : SUCCESS in 1h 41m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f5adcb0289148b8b6ab2912e2acada9 : SUCCESS in 56m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c7242220e4ab4c318874c22416cd5111 : SUCCESS in 14m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/adbd3127cf774fa586a947a5935b47df : SUCCESS in 5m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/310ab507c79042d59a5667b9e04c8bc3 : FAILURE in 5m 49s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/034075e5e6534e9597b42359a3a899c6 : SUCCESS in 6m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/12a43fca3a074bbc9ca380be47b87169 : SUCCESS in 5m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0ac9169e64224ebf856130f524b2e4ce : SUCCESS in 39m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a5206f697c664a16a6b242e8425318a2 : FAILURE in 5m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8a4008139db04b938480a96b880d96b7 : SUCCESS in 1h 08m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3a60c501d25d4d6c89956da4a233615d : FAILURE in 1h 23m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/59753d048db54f26a8ccac492fcce4e7 : SUCCESS in 1h 38m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/593bc355390542c5a58774039aa42896 : SUCCESS in 1h 02m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/85f16fbf5a424d1bb49be79a911d6c14 : SUCCESS in 1h 31m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/53a79c02798d4f58a78cbd9e7a2aa102 : SUCCESS in 38m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/43c994567a8243f78297a1b2886291be : SUCCESS in 1h 31m 58s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/94a5d6d622e04ba6bf5368312f1cf88c : SUCCESS in 2h 10m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f17ccef7d3444ea0a0bf2c9e37652506 : SUCCESS in 49m 43s","accounts_in_message":[],"_revision_number":5},{"id":"1c6aa50197363e2a44808b30dc78b5d3ee4baef4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-11-22 07:01:48.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/850840/5/upstream-check/cinder-tempest-plugin-ontap-fcp/59e3531/ : RETRY_LIMIT in 8m 13s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/850840/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/7136f0a/ : SUCCESS in 43m 35s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/850840/5/upstream-check/cinder-tempest-plugin-ontap-nfs/3cf0bee/ : FAILURE in 32m 34s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"af910db36d2572a5cdb4a01d000e22b47adc6eb1","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-11-22 07:07:56.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/850840/5/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/6697474/ : SUCCESS in 49m 32s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"0ea2c87236309ff277217e62c5068761fc65b14a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-11-22 07:16:39.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/40/850840/5/ : SUCCESS in 1h 59m 19s","accounts_in_message":[],"_revision_number":5},{"id":"0758e3b27079af89008a74178d78ba5fd3cd5f37","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-22 07:35:32.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-40-850840-5/ns5_nfs : SUCCESS in 3h 31m 24s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-40-850840-5/ns5_iscsi : SUCCESS in 3h 35m 55s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"274671e317b61a208e98fabd87dcf804934d46fd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-22 07:43:39.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/40/850840/5/check/DellEMC_Unity_FC/78fd38e/DellEMC_Unity_FC/2018 : SUCCESS in 59m 12s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/40/850840/5/check/DellEMC_Unity_iSCSI/b67a3f4/DellEMC_Unity_iSCSI/1865 : SUCCESS in 2h 15m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"4130f260b283f155dfeff827f5ea2a1187179890","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-11-23 11:02:56.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/40/850840/5/20221123132312/NEC-ISCSI : SUCCESS in 3h 31m 52s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/40/850840/5/20221123132312/NEC-FC : SUCCESS in 6h 39m 41s","accounts_in_message":[],"_revision_number":5},{"id":"38bd0e18417e9d040dee51968cd4a866af377b5f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-11-23 12:55:11.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/40/850840/5/20221123155547/NEC-V-ISCSI : SUCCESS in 2h 33m 11s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/40/850840/5/20221123155547/NEC-V-FC : SUCCESS in 5h 59m 20s","accounts_in_message":[],"_revision_number":5},{"id":"87005370a4825246ec64917434c527d01bcaee36","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-03-22 03:20:53.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6b4617acfd9fedbc0e2afdac1a4ab848bc24687a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-03-24 02:39:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"32419875e2c5f95927f1b9f7a7b7070ee2227c37","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-24 02:59:17.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/40/850840/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/469aee7/ : FAILURE in 12m 41s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/40/850840/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/e6e9d8e/ : FAILURE in 11m 45s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/40/850840/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/89d07f2/ : FAILURE in 12m 18s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/40/850840/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/b969afc/ : FAILURE in 12m 08s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/40/850840/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/681a3d9/ : FAILURE in 12m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"66feadbb657458223a3f6d0272c8ce7a43962f8a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-24 03:08:51.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/850840 : FAILURE in 27m 36s","accounts_in_message":[],"_revision_number":6},{"id":"e754dbb7592fc667742386decf46390e95dcfdb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 03:23:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1cbd7ba7a7141f08fdb27ea3d3178e5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/59f497de65c547d1a2bd5cc1f64beb82 : FAILURE in 26m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c5cb473ea16148489bf1f75422278b27 : SUCCESS in 21m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8bbb5468c248f524d628c5cf1b864a93b208ee1e","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-24 03:36:20.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/ded7d2a988164ca2b914aeab4b494710 : FAILURE in 55m 57s","accounts_in_message":[],"_revision_number":6},{"id":"c39bea661c5bdac7830fb635ec4cb9e947ea5185","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-24 03:56:24.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/15ce2ebe6f864fc899b63330309488d3 : SUCCESS in 1h 15m 04s","accounts_in_message":[],"_revision_number":6},{"id":"d487c07b65250fdfc5eb61ccdf421032ddf64ea4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-24 04:04:10.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/40/850840/6/check/iscsi-driver-alletra-6k/c9ddada/c9ddada.html : FAILURE in 1h 23m 32s","accounts_in_message":[],"_revision_number":6},{"id":"ffc1c9b14077391a3e5613bbd4f927759950c45c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-24 04:38:46.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder finger://openstack-ci-fc/7a55db5ac74445fc92ac12b89d9a97c6 : RETRY_LIMIT in 18m 58s","accounts_in_message":[],"_revision_number":6},{"id":"f882616a9d4cfca02e73ca245c6a6c32d9b0faa0","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-03-24 04:43:35.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/6/check/fujitsu-eternus-dx-iscsi/66bc63f : FAILURE in 1h 58m 07s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/6/check/fujitsu-eternus-dx-fc/b98b624 : SUCCESS in 2h 03m 17s","accounts_in_message":[],"_revision_number":6},{"id":"d3dfb09fc787481e5cc80ec08e880b30bc559e62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 04:52:45.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5e219e222774e7e9dd358817f514ac2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a824dc85390c45c18fa5b7e5156fb9af : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/885997af18ef484fa3fcdd7e39e5a96e : SUCCESS in 12m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c475453dc52148049e07933e5a313238 : SUCCESS in 16m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/938bf5c90369420abf95f09f20ab66f5 : SUCCESS in 13m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/990dba7a94d74e4da2d46a1c9cf56a94 : SUCCESS in 1h 14m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d5d134492c444e0586457b5e7a5a1c6f : SUCCESS in 1h 36m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6039e7cca0934f0cbed84323dc44767e : SUCCESS in 1h 05m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fafc02c297c44e5bdec1b1b5f09751a : SUCCESS in 6m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0876f5c2977141a49c0e266dc68e0cc1 : SUCCESS in 14m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e17c2cad120d4bdbaaad887144a12ceb : SUCCESS in 5m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/75bc336765554c6cb712bd761ca38616 : FAILURE in 6m 37s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9180c425ae2942d4bed434fa74d84322 : SUCCESS in 5m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aa95195d4bd14887a8fabef9b1fa870f : SUCCESS in 5m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5e5df81254e84e45b427840155bccaac : SUCCESS in 35m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5113511db9de43539af9a2f9c6885d84 : FAILURE in 7m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6fcf8eea49214f9997089479b585d344 : SUCCESS in 1h 01m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4119234674804356845fc08ec4a009a3 : FAILURE in 1h 03m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f760b9a0916e4ad5b696d392edaebf30 : SUCCESS in 1h 44m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0d9067c8e2664a07aaa21c870681fae4 : FAILURE in 52m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8e961cc89e9347d2878a7140c6f0fbef : SUCCESS in 1h 29m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/452bbc1cca534b2980184860a057b4ff : SUCCESS in 40m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/17fcbbf979aa432a80c7f505563ede1f : SUCCESS in 1h 31m 52s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6c2dfb6b7d9445dd8b972efaf86034e7 : SUCCESS in 1h 50m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53a4f915451f4c55b727502b2780e3ec : SUCCESS in 46m 00s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e506ac9a4724bb0a73edd51c7f1279e : SUCCESS in 1h 16m 53s","accounts_in_message":[],"_revision_number":6},{"id":"166b5dffd48afd1c20bd9013e033d16d48f64345","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-03-24 05:16:58.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"42d8b54d60b0abac9a15dfee46aa9228c62e3053","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-03-24 05:17:52.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c622dde7289881a8cb113016b18da59f1991467c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 05:29:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fb65af931cd4118917ddcb0978989a4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/064178a7721545e2b4b612d24a8d8e52 : SUCCESS in 11m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/76ece8d2777e44b8b28897cf917f3e18 : SUCCESS in 9m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5e1ce8c5fa3a9eed8c89fc795ad1da0d67b0b022","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-24 06:10:03.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-40-850840-6/ns5_nfs : SUCCESS in 3h 26m 54s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-40-850840-6/ns5_iscsi : SUCCESS in 3h 30m 0s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"421c1d766747a5638fcb66c547e4d08dc7568a1b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-24 06:11:34.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/40/850840/6/check/tempest-dsvm-full-focal-py3/188b62b/ : SUCCESS in 1h 37m 27s\n- devstack-plugin-ceph-dsvm-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/40/850840/6/check/devstack-plugin-ceph-dsvm-focal-py3/0679df6/ : SUCCESS in 1h 47m 12s","accounts_in_message":[],"_revision_number":6},{"id":"3b2939479f1af6ba4cd31d70f83c648c58c86362","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-24 06:15:22.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/8a25ddec99fe4b668f6c0f3eebd0a988 : FAILURE in 57m 38s","accounts_in_message":[],"_revision_number":6},{"id":"9ab028648d6f23acee676192243d693656c4dbb2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-24 06:43:31.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/11a15efe158d4b13a95524a1d6c6483d : DISK_FULL in 1h 24m 48s","accounts_in_message":[],"_revision_number":6},{"id":"989f9c4d6c0b4d3bb71269451480fad615342246","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-03-24 06:55:49.000000000","message":"Patch Set 6:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/6/check/fujitsu-eternus-dx-iscsi/ebf81e4 : SUCCESS in 1h 37m 22s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/6/check/fujitsu-eternus-dx-fc/e179f89 : SUCCESS in 1h 32m 13s","accounts_in_message":[],"_revision_number":6},{"id":"83bfe2686137fd0b1b829a41b06ad9b1d2c5da60","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-24 07:29:37.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/40/850840/6/check/DellEMC_Unity_FC/14feabd/DellEMC_Unity_FC/2564 : SUCCESS in 1h 49m 52s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/40/850840/6/check/DellEMC_Unity_iSCSI/b21e2dd/DellEMC_Unity_iSCSI/2411 : SUCCESS in 2h 13m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"9245909571d7111b74c13d8e01a887f6fac334bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 07:35:41.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/369c99a8303b4950b690d528a70c8231\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42f8135cf47c42c4b34d8c1282615ef0 : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6642ff832d143bd9bbec297535825f3 : SUCCESS in 12m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f7870a68bef40bc8a64aae9dcf130f6 : SUCCESS in 11m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fc743462fc04f2caae72600c02bc6a3 : SUCCESS in 12m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/d60c787e0d8844d59498f34c2db7da90 : SUCCESS in 47m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/13e2a77dc3e3402ab71cc20fb5416e51 : FAILURE in 1h 14m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03279e52f06e4977870ebca6776b45d8 : SUCCESS in 50m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e958e9d5700b4ccc8efc507d390c8a53 : SUCCESS in 8m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c0a22d17ce65406c8947a8dbe4eaa18a : SUCCESS in 14m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3a4001e45ab24ca5beacc99996a91fb2 : SUCCESS in 5m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f7ac0e1d48184b4d9688ef6a1d6698b1 : FAILURE in 6m 22s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9d46d2463aaa48ab9fde924879d5a603 : SUCCESS in 6m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e8d39ca7ce454035bf5f359bf3f3be42 : SUCCESS in 5m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/60a68024fb59472da2e676f966e8bc6f : SUCCESS in 40m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ad9537bcf8b148babc0bddc9c5e31e07 : FAILURE in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cd354f9786ef4558880aecae4c5d476e : TIMED_OUT in 2h 05m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fc828f7debf54df8a8327177a2bded67 : FAILURE in 1h 22m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5e56757e607a4d929c31fd2bb086e1ea : SUCCESS in 1h 19m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fdc9f1b0d61a45c4a988ff0b4a71ce82 : SUCCESS in 1h 06m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4dc5a787f0c54aa4b7a9f290abec6683 : SUCCESS in 1h 24m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bb5f3a43ddcc49468303a747031930fb : SUCCESS in 33m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/aa116b6ba9ff4eaf92abd100b7b5022d : SUCCESS in 1h 18m 39s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2ccf2b02fce14ddfa0681e72e09bb2d5 : SUCCESS in 2h 14m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/833d6b1109e4463ba3e50f5fd5de0c55 : SUCCESS in 48m 06s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80a66421d1384deea6c4a07fcbdb88fe : SUCCESS in 1h 13m 10s","accounts_in_message":[],"_revision_number":6},{"id":"9c4985cbccd620d3bb36c4dd8d0f6077b22dc8e2","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-03-24 07:41:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b040293bfb850c908a67a0607836f375e323ce4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 07:51:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bda4db7b976e47fd83bc4d9e3ade3aa9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2b9cd73b80a247d6abca5558ebb83c27 : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/14177773647f4d38af464cb776b1aac1 : SUCCESS in 8m 58s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"61ffc9b4e48da34db02f0c0de90a0701a26e0ae0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-24 08:01:27.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/40/850840/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2fd47a2/ : FAILURE in 12m 13s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/40/850840/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/97c2a03/ : FAILURE in 12m 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/40/850840/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/07920a0/ : FAILURE in 12m 30s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/40/850840/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/2f52a23/ : FAILURE in 11m 48s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/40/850840/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/6f403b5/ : FAILURE in 12m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"65dccd47e0ade19846eb7e372b47f5709a9f1836","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-24 08:41:24.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/c25d25b19e924d72b1a50ad28eb93734 : FAILURE in 58m 56s","accounts_in_message":[],"_revision_number":6},{"id":"e0f72fc7116ba6a3f53dcea08a776bd59257c6eb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-24 08:54:37.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/00a9cf2ef8d04f97983f7ed8993e659a : SUCCESS in 1h 11m 08s","accounts_in_message":[],"_revision_number":6},{"id":"2049e505a60067439cd565d5858218ddff4194fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 09:49:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a87f7c268ef48fe8a292b4f57ef882a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c60c84d4462d4b658be3817c5341de6e : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b56e1290a8045308de64cf292906503 : SUCCESS in 10m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bc2b90892664f6bbb122201a077d889 : SUCCESS in 11m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9f17c3121654ce6ad4044b3f6bef8e0 : SUCCESS in 10m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/167820148dc6476c86f5227d1316db2e : SUCCESS in 58m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/41331456c9eb4745a20c2a2dcb4a00b6 : SUCCESS in 1h 24m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e1398c099694b85a7298c8673a81c11 : SUCCESS in 58m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b93416a318dd47799be0c660355fc022 : SUCCESS in 9m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ca3324a0db7540a58174ce6c86b31c81 : SUCCESS in 15m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/99c9c9f291454c08af93f8be1cdddf1d : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b940de1cfdf04288976b12dd1bc7ca9e : FAILURE in 6m 10s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/79a8a49160814d798675f3d6607ebc5c : SUCCESS in 6m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7edf9c8d6f2b42049a9ebcfdcdafccad : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/36bbdca48e764a02ac2c4fdc61b92c4a : SUCCESS in 37m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f0b687ea52674daba56bd7de14c2e49c : FAILURE in 6m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e955775644f746adbfa4ed2bf274f272 : SUCCESS in 1h 18m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a1ed8fc5785d4e1db0b5d89aedc269fe : FAILURE in 1h 26m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/09c36515676b42ff9f400fc1c7b05556 : SUCCESS in 1h 57m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2322781a2d88448996010fc3f2abca88 : SUCCESS in 1h 01m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3d720ea4dae74503a3c2610051904d19 : SUCCESS in 1h 31m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fd8899a75af74c48b19ed03936cf277d : SUCCESS in 47m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e2b7e5d2b2d94997a351984120dfdae2 : SUCCESS in 1h 04m 28s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/caf19d6354d346db93e852c0c89ccedc : SUCCESS in 2h 07m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/532074de3b1045c6b2fa94bed7dcf836 : SUCCESS in 39m 20s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/025a2d8aed5742099cb8fb6a67381539 : SUCCESS in 1h 19m 28s","accounts_in_message":[],"_revision_number":6},{"id":"da7f3d65438e91267131c7dd25c17a95ec6770ab","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-03-24 09:51:09.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"c801dc13bc72f7934262f1f98a3e8af2e6133cdb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-03-24 09:58:11.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_03_2023_09_50_04\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"7b4d1451bf450ff4f7098ce8e83a6410f4476785","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-03-24 10:26:19.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/6/check/fujitsu-eternus-dx-iscsi/f53f054 : FAILURE in 2h 42m 34s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/6/check/fujitsu-eternus-dx-fc/e0b3d6b : SUCCESS in 2h 43m 59s","accounts_in_message":[],"_revision_number":6},{"id":"502467120486f79b41f497343d26d113fcdb5bca","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-03-24 10:49:52.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/850840/6/upstream-check/cinder-tempest-plugin-ontap-fcp/10abe06/ : SUCCESS in 59m 07s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/850840/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/0fa2e10/ : SUCCESS in 44m 29s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/850840/6/upstream-check/cinder-tempest-plugin-ontap-nfs/8a0fcd5/ : SUCCESS in 42m 51s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/850840/6/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/2ed319d/ : FAILURE in 43m 28s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"9fb69035c33fd23b23a2f80525458a12fcbec3df","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-03-24 11:24:32.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/40/850840/6/check/DellEMC_PowerFlex/0ffe79c/DellEMC_PowerFlex/1782 : SUCCESS in 1h 34m 15s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"540c7bf51b4d4052af0529d6560cd6181ba9925f","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2023-03-24 11:46:18.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/40/850840/6/check/DellEMC_SC_FC/d00a3e9/DellEMC_SC_FC/1248 : SUCCESS in 1h 24m 18s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/40/850840/6/check/DellEMC_SC_iSCSI/dc34179/DellEMC_SC_iSCSI/1228 : SUCCESS in 1h 56m 02s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":6},{"id":"6562cb816f9b985d65b79e2d996d30eca0942427","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-24 11:46:56.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/40/850840/6/check/tempest-dsvm-full-focal-py3/dc5de89/ : SUCCESS in 1h 54m 39s\n- devstack-plugin-ceph-dsvm-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/40/850840/6/check/devstack-plugin-ceph-dsvm-focal-py3/1379884/ : SUCCESS in 2h 03m 53s","accounts_in_message":[],"_revision_number":6},{"id":"25242e6e2c3ea9a7278d22e19ec7a2771ad92494","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-03-24 14:31:43.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/850840-6/PVME-iSCSI/ : SUCCESS in 1:44:07\n- PVME-FC https://os-logs.seagate.com/dell-ci/850840-6/PVME-FC/ : FAILURE in 1:31:52","accounts_in_message":[],"_revision_number":6},{"id":"a37f4b72105da66e3f899215456113d130124855","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-03-24 15:28:40.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/850840-6/Seagate-iSCSI/ : SUCCESS in 1:47:35\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/850840-6/Seagate-FC/ : SUCCESS in 1:32:03","accounts_in_message":[],"_revision_number":6},{"id":"0a47d0cae3429f894a4cf04902944c045f5a8b44","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-24 15:29:18.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/40/850840/6/check/DellEMC_Unity_FC/37fb3ec/DellEMC_Unity_FC/2568 : SUCCESS in 1h 43m 55s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/40/850840/6/check/DellEMC_Unity_iSCSI/dfb7692/DellEMC_Unity_iSCSI/2415 : SUCCESS in 1h 59m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"ced7211cf3be48fc4b849b3f9718e0f0a0886e6f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-03-26 08:53:25.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/40/850840/6/20230326122318/NEC-V-FC : SUCCESS in 2h 39m 52s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/40/850840/6/20230326122318/NEC-V-ISCSI : SUCCESS in 5h 30m 04s","accounts_in_message":[],"_revision_number":6},{"id":"3c36bf0aff3c78c02b60faaaad893f97590abf10","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-03-27 01:32:24.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/40/850840/6/ : POST_FAILURE in 1h 10m 15s","accounts_in_message":[],"_revision_number":6},{"id":"6def1fc1243df67f9cb846baa2319d5f5a3d6085","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-03-27 05:33:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cc562ec8a39e0966e2cccc3dc2f2135559917224","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-03-27 07:14:12.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/6/check/fujitsu-eternus-dx-iscsi/764f81f : SUCCESS in 1h 39m 16s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/6/check/fujitsu-eternus-dx-fc/236068b : FAILURE in 2m 19s","accounts_in_message":[],"_revision_number":6},{"id":"0f4048eb99a8c09d1d9504f547f21ed44d256a6f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-03-28 00:27:14.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-40-850840-6/FC/ : SUCCESS in 2h 21m 29s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-40-850840-6/iSCSI/ : SUCCESS in 2h 36m 36s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"d16f6e7ece76f5fdd7c0f7937c2dc19bda2f07c0","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-03-28 01:11:07.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6e406f1febd0b68b3eef9fa62416a05a68bb3c33","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-20 07:20:15.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/40/850840/6/check/iscsi-driver-nimble/7ca6915/7ca6915.html : FAILURE in 17m 53s","accounts_in_message":[],"_revision_number":6},{"id":"0eba1443bceda1add8737fae99ad85d95301b1b4","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-04-30 21:34:37.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/40/850840/6/20230430233424/NEC-ISCSI : SUCCESS in 3h 23m 16s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/40/850840/6/20230430233424/NEC-FC : SUCCESS in 7h 00m 09s","accounts_in_message":[],"_revision_number":6},{"id":"854b60e7d740df6586f07262534f0722347e4864","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-05-23 02:23:26.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ea5bb1c6b1ae6bef84de43f11e481fd82c9cc345","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-05-29 03:36:42.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c2e4da57004cc595c3e64276cfc15a9a1806330a","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-05-29 05:30:30.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"050b7ba97cb9c321d658be0bb31a80227dcbecab","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-05-29 06:13:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a574dbf487e708c0e6afdd3850f0cccf3cba3a2d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-05-29 07:22:49.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/6/check/fujitsu-eternus-dx-iscsi/2b41ba1 : FAILURE in 11s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/6/check/fujitsu-eternus-dx-fc/0083f16 : SUCCESS in 1h 51m 27s","accounts_in_message":[],"_revision_number":6},{"id":"92e787dde337272bc06c1eca7e32217c45d1c512","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-05-29 07:40:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5acc91cbf08a2bd44332de243940f9d1c2b1b39a","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-05-29 08:01:40.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/6/check/fujitsu-eternus-dx-iscsi/089f516 : ABORTED in 19m 00s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/6/check/fujitsu-eternus-dx-fc/a2375c9 : ABORTED in 18m 57s","accounts_in_message":[],"_revision_number":6},{"id":"454fb39a575e0c7e1953a84954e893f366598b24","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-05-29 08:27:19.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5be7cb7712ef543601735ca8d632be1c967acaf3","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-05-29 10:23:29.000000000","message":"Patch Set 6:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/6/check/fujitsu-eternus-dx-iscsi/011ca00 : SUCCESS in 1h 41m 07s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/6/check/fujitsu-eternus-dx-fc/bb4c1a6 : SUCCESS in 1h 55m 29s","accounts_in_message":[],"_revision_number":6},{"id":"72fb38ef7cb02874650ad526b3f85a9e827b867e","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-05-29 10:42:36.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"589d6c67fef2a8065db42c47556f358bc991bd4e","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-05-30 10:37:28.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f9c6edea51efe2674ac255a1b8c736036b2fd032","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-05-31 22:00:08.000000000","message":"Patch Set 6:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/850840/6/fc : SUCCESS in 0h 44m 56s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/850840/6/iscsi : SUCCESS in 0h 53m 20s","accounts_in_message":[],"_revision_number":6},{"id":"1b20e09f68ce44f300785085b8474dd598617412","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-06-07 08:40:14.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"3e0a70828fef5e2a237583a9d805a652fa102650","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-07 08:41:03.000000000","message":"Patch Set 7:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":7},{"id":"471f4817c6f925504e09d2a88ce8e06c6452b20b","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-06-07 08:46:47.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c8a3a699fc57d0fb52500836718bc72a4077d9b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 09:04:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4af3db19e634e51a7aba9f1e2339c6d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/66d4d4422bbd4c9d98f9cd81836d5456 : FAILURE in 21m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cca3cd0404f14756a66c96fff856ab94 : FAILURE in 9m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bf5eea4fc85b45cf997446f0f8ec74e1 : FAILURE in 22m 07s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8b661e583be20bee6f31ea899a2eafe1a83e3eb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-06-07 09:17:50.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"46eceed3be551224bb8360a049ee41e85865b395","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-07 09:18:40.000000000","message":"Patch Set 8:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":8},{"id":"6a4710032891f57f34ba6c2a51ab675e11c902dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 09:40:43.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4b24341ed7f4ff1a1ef99af803ad267\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7db34bb9351b4d1f95ccec7addec1d31 : SUCCESS in 21m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b510c1eb4ee142edab57fbd1a897a243 : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dd1e6553f2fc465eb9b59be49750c59d : SUCCESS in 9m 28s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"05f338f23a47792143f38c2f43f7eec72a9c8c8b","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-06-07 10:06:36.000000000","message":"Patch Set 8: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"85c290604dadd82535412cdb8284a2ab9cd90d3c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-07 10:47:59.000000000","message":"Patch Set 8:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/ab5c4d67840d4e10a4e7b2af72210e14\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/7fc5c5fa751749b688af64c9737f17ce : SUCCESS in 1h 27m 15s","accounts_in_message":[],"_revision_number":8},{"id":"7dc6e69105bce6680de4ea7d5387b226fd666a60","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-07 11:07:46.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/40/850840/8/check/iscsi-driver-alletra-6k/159d964/159d964.html : FAILURE in 1h 44m 40s","accounts_in_message":[],"_revision_number":8},{"id":"d4befaee2693d038b84c522eb199505887d97059","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-06-07 11:28:03.000000000","message":"Patch Set 8:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/40/850840/8/check/fujitsu-eternus-dx-iscsi/e52514a : SUCCESS in 2h 03m 07s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/40/850840/8/check/fujitsu-eternus-dx-fc/3f177e4 : SUCCESS in 2h 08m 28s","accounts_in_message":[],"_revision_number":8},{"id":"37623f0f506f875367cce9b79fe5dc0db09c7e87","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-07 11:32:34.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/40/850840/8/check/DellEMC_Unity_FC/dfa26c5/DellEMC_Unity_FC/236 : SUCCESS in 1h 43m 06s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/40/850840/8/check/DellEMC_Unity_iSCSI/afa7e85/DellEMC_Unity_iSCSI/236 : SUCCESS in 1h 52m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"4bfda82d0f03a68cfdf9adeb055817b6cb6c407c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-07 11:39:46.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-40-850840-7/FC/ : SUCCESS in 2h 59m 11s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-40-850840-7/iSCSI/ : SUCCESS in 2h 11m 14s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"1c593a700bc6297400d4cf97ba0727d55127a015","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 11:41:13.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/908ef772ea764f318a01d9f8a8e43445\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c168e8e48c0495fb350a3e4517e873e : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/111d718d54374cfb865aee22c4d8ed5e : SUCCESS in 12m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4bf7d33e9d544fb9b300213abe960b1c : SUCCESS in 12m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86d0b328810b483ab45505f9aa685850 : SUCCESS in 12m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14bb57c4408d483391c85011a3124f68 : SUCCESS in 13m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/e3dc84cf3000491299b2f2d172b2d456 : SUCCESS in 52m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/83256d872d5e439a962f95f39643cc0b : SUCCESS in 1h 45m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b83c39b57474680bc8d59190d9ac7ec : SUCCESS in 59m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15655b9169e44d5499eb1cb42e23c978 : SUCCESS in 8m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d66872b5dd444706973585fb60d5a78a : SUCCESS in 16m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d86afb8e0df545a49158429309aa140d : SUCCESS in 6m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/11afa67b9e97479d96c7021103b53c5e : FAILURE in 5m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c034e101fe514721b4223879c99a77ee : SUCCESS in 6m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5fe461f0d6c64a6c82a765f56fe915ac : SUCCESS in 6m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4349a4c793024d6b91f8771cdbb5ad52 : SUCCESS in 43m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9a20ce87a7bd42aeb2573dd41358881d : SUCCESS in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6731503748b04612a7205f9728b0d11d : SUCCESS in 2h 05m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4e76683e32cc4a558e773669ceda124b : FAILURE in 1h 56m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/75d258b1b23148ac86431fb656698cec : SUCCESS in 2h 12m 09s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7916a08559654c0dadddd14760cfc9d2 : SUCCESS in 2h 22m 30s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fc7fc6ba64354093a64913aad412adb7 : SUCCESS in 48m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c7637c80f9c045ad9b7badb42994ed19 : SUCCESS in 1h 31m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8da011ad9024466d967263646381a634 : SUCCESS in 44m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4a9db4f0bed34c6d99b9b90be5088e2c : SUCCESS in 1h 25m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6e0a3fabb7ba42a1a5962af3bcad840f : SUCCESS in 1h 32m 46s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3f13bb4f67a9491f961ca68d852fd041 : SUCCESS in 2h 11m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1838988ad72f49d69063354d5637e377 : SUCCESS in 1h 00m 30s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/5910cf90721742289a084efcc7f9096e : SUCCESS in 12m 27s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4fcf946968f089201d07ad6a6ce74e0605e0bc41","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-06-07 11:48:14.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/07_06_2023_11_41_21\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"04231f1aca50cb9e610745c7680c4c89e35c8ef5","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-07 11:56:27.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-40-850840-8/FC/ : SUCCESS in 2h 38m 16s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-40-850840-8/iSCSI/ : SUCCESS in 2h 9m 3s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"2dc54506153b5a80a835b561d704697c6a877d05","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-06-07 13:06:43.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/40/850840/8/check/DellEMC_PowerStore_iSCSI/b07f50b/DellEMC_PowerStore_iSCSI/83 : SUCCESS in 53m 35s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/40/850840/8/check/DellEMC_PowerStore_FC/471783b/DellEMC_PowerStore_FC/88 : SUCCESS in 1h 23m 24s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/40/850840/8/check/DellEMC_PowerStore_NFS/10ad978/DellEMC_PowerStore_NFS/82 : SUCCESS in 58m 53s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/40/850840/8/check/DellEMC_PowerStore_NVMeTCP/8cd0a79/DellEMC_PowerStore_NVMeTCP/83 : SUCCESS in 1h 06m 03s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"b255fccf2380e2b919bda9107dcf89170c5cde80","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-06-07 13:24:09.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/850840/8/upstream-check/cinder-tempest-plugin-ontap-fcp/7d2e416/ : RETRY_LIMIT in 9m 46s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/850840/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/869de0d/ : FAILURE in 51m 33s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/850840/8/upstream-check/cinder-tempest-plugin-ontap-nfs/ccd3b24/ : SUCCESS in 1h 41m 42s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/850840/8/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/c4bb366/ : FAILURE in 1h 38m 17s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"63dd5106ac57e077bf4c7e0263a93850821f27c8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-07 14:16:38.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/40/850840/8/check/DellEMC_PowerFlex/fe37439/DellEMC_PowerFlex/131 : SUCCESS in 1h 39m 44s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"0def6469e9171990564c500ea399b50f9119257c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-07 17:25:57.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/40/850840/8/check/DellEMC_Unity_FC/7fa8d4a/DellEMC_Unity_FC/239 : SUCCESS in 1h 54m 30s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/40/850840/8/check/DellEMC_Unity_iSCSI/ffe8767/DellEMC_Unity_iSCSI/239 : SUCCESS in 2h 03m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"f1807ffda2b6fcd6bb79c7baeb4cb73d4be994b7","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2023-06-07 18:01:39.000000000","message":"Patch Set 8:\n\nBuild failed.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/850840-8/Lenovo-iSCSI/ : FAILURE in 36:17.29\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/850840-8/Lenovo-FC/ : FAILURE in 36:06.30","accounts_in_message":[],"_revision_number":8},{"id":"b6e70a1c953261a1ba1cd1cca25b1538133c6cc1","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-06-08 00:10:36.000000000","message":"Patch Set 8:\n\nBuild failed.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/850840-8/HPMSA-iSCSI/ : FAILURE in 1:45:50\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/850840-8/HPMSA-FC/ : FAILURE in 1:26:02","accounts_in_message":[],"_revision_number":8},{"id":"ff29676e6730a04705b48433c18cd67672be4753","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-06-08 02:31:01.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/40/850840/8/check/DellEMC_PowerMAX_FC/eda1834/DellEMC_PowerMAX_FC/106 : SUCCESS in 7h 21m 43s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/40/850840/8/check/DellEMC_PowerMAX_iSCSI/ea68e3d/DellEMC_PowerMAX_iSCSI/101 : FAILURE in 6h 34m 45s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"121dcbb0df2fc1e335bbcd0849d8112c20875450","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-06-08 02:54:12.000000000","message":"Patch Set 8:\n\nBuild failed.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/850840-8/PVME-iSCSI/ : FAILURE in 1:58:32\n- PVME-FC https://os-logs.seagate.com/dell-ci/850840-8/PVME-FC/ : FAILURE in 1:37:25","accounts_in_message":[],"_revision_number":8},{"id":"dd32e7f7086e1fa99b8360dd312f8f624c3616d2","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-08 05:00:09.000000000","message":"Patch Set 7:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/850840/7/fc : SUCCESS in 0h 34m 57s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/850840/7/iscsi : SUCCESS in 0h 47m 26s","accounts_in_message":[],"_revision_number":7},{"id":"ad3328e28089bcce3d720668be83f66fce193d84","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-08 08:00:08.000000000","message":"Patch Set 8:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/850840/8/fc : SUCCESS in 0h 41m 27s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/850840/8/iscsi : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":8},{"id":"1573956582fa45fa69de0e5859c1c3e15c2b10c5","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-08 12:00:08.000000000","message":"Patch Set 8:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/850840/8/fc : SUCCESS in 0h 41m 27s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/850840/8/iscsi : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":8},{"id":"9e1dc199817195b223d69235a656e1db473d85cd","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-08 16:00:07.000000000","message":"Patch Set 8:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/850840/8/fc : SUCCESS in 0h 41m 27s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/850840/8/iscsi : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":8},{"id":"12e3602a66c9d3bba224b2bcd0681d1e9a787751","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-08 18:00:08.000000000","message":"Patch Set 8:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/850840/8/fc : SUCCESS in 0h 41m 27s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/850840/8/iscsi : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":8},{"id":"c3d3d94d17dd282ef9e55871652888922f4da623","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-08 19:00:07.000000000","message":"Patch Set 8:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/850840/8/fc : SUCCESS in 0h 41m 27s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/40/850840/8/iscsi : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":8},{"id":"fa1858b6bad58dcc5324d9ffb99e69dcc7e3104f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-06-15 20:57:06.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/40/850840/8/20230615215118/NEC-V-ISCSI : FAILURE in 5h 37m 58s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/40/850840/8/20230615215118/NEC-V-FC : FAILURE in 8h 05m 45s","accounts_in_message":[],"_revision_number":8},{"id":"867d01eef7b7d06a8b5d016daee0f8eb7e68a8e1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-06-16 22:53:43.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/40/850840/8/20230616234725/NEC-ISCSI : FAILURE in 8h 03m 30s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/40/850840/8/20230616234725/NEC-FC : FAILURE in 8h 06m 14s","accounts_in_message":[],"_revision_number":8},{"id":"0188157f08cacd6a449224965ef86d4d9c50eeb2","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-07-05 03:52:33.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"11969252042761dbab99294fb73036ddacd2847c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-09-15 01:16:10.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"469a4989dbd6e6183eb7a2b07827ce617010b5ed","revisions":{"2c745216fdfb2f2bf8d9379ebde1351548d5b3ea":{"kind":"REWORK","_number":1,"created":"2022-07-25 08:00:16.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/40/850840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/850840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/850840/1"}}},"commit":{"parents":[{"commit":"f8fa57a8b974f30cdda8367ebd470279871ea136","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8fa57a8b974f30cdda8367ebd470279871ea136"}]}],"author":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-25 08:03:55.000000000","tz":-240},"committer":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-25 08:03:55.000000000","tz":-240},"subject":"Fujitsu Driver: Add volume migration support","message":"Fujitsu Driver: Add volume migration support\n\nAdded support for volume migration in the Fujitsu driver.\n\nAdded below function:\n\nupdate_migrated_volume\n\nChange-Id: Ifbb1d7d5271ef62cc12f4de8d916661bdba67640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c745216fdfb2f2bf8d9379ebde1351548d5b3ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c745216fdfb2f2bf8d9379ebde1351548d5b3ea"}]},"branch":"refs/heads/master"},"8503a1a2c081833cc6dbdaf7178caf4471de9f0e":{"kind":"REWORK","_number":2,"created":"2022-07-26 01:46:00.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/40/850840/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/850840/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/850840/2"}}},"commit":{"parents":[{"commit":"f8fa57a8b974f30cdda8367ebd470279871ea136","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8fa57a8b974f30cdda8367ebd470279871ea136"}]}],"author":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-25 08:03:55.000000000","tz":-240},"committer":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-26 01:48:57.000000000","tz":-240},"subject":"Fujitsu Driver: Add volume migration support","message":"Fujitsu Driver: Add volume migration support\n\nAdded support for volume migration in the Fujitsu driver.\n\nAdded below function:\nupdate_migrated_volume\n\nChange-Id: Ifbb1d7d5271ef62cc12f4de8d916661bdba67640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8503a1a2c081833cc6dbdaf7178caf4471de9f0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8503a1a2c081833cc6dbdaf7178caf4471de9f0e"}]},"branch":"refs/heads/master"},"b3c590e1730ca87fd494c5ea76fe58794734a3ff":{"kind":"REWORK","_number":3,"created":"2022-07-26 06:44:46.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/40/850840/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/850840/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/850840/3"}}},"commit":{"parents":[{"commit":"f8fa57a8b974f30cdda8367ebd470279871ea136","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8fa57a8b974f30cdda8367ebd470279871ea136"}]}],"author":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-25 08:03:55.000000000","tz":-240},"committer":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-26 06:48:10.000000000","tz":-240},"subject":"Fujitsu Driver: Add volume migration support","message":"Fujitsu Driver: Add volume migration support\n\nAdded support for volume migration in the Fujitsu driver.\n\nAdded below function:\nupdate_migrated_volume\n\nChange-Id: Ifbb1d7d5271ef62cc12f4de8d916661bdba67640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b3c590e1730ca87fd494c5ea76fe58794734a3ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b3c590e1730ca87fd494c5ea76fe58794734a3ff"}]},"branch":"refs/heads/master"},"965397fbf7f533cbad978d649caebf2083a248f7":{"kind":"REWORK","_number":4,"created":"2022-11-21 00:57:03.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/40/850840/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/850840/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/850840/4"}}},"commit":{"parents":[{"commit":"f8fa57a8b974f30cdda8367ebd470279871ea136","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8fa57a8b974f30cdda8367ebd470279871ea136"}]}],"author":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-25 08:03:55.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2022-11-21 01:01:43.000000000","tz":-300},"subject":"Fujitsu Driver: Add volume migration support","message":"Fujitsu Driver: Add volume migration support\n\nAdded support for volume migration in the Fujitsu driver.\n\nAdded below function:\nupdate_migrated_volume\n\nChange-Id: Ifbb1d7d5271ef62cc12f4de8d916661bdba67640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/965397fbf7f533cbad978d649caebf2083a248f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/965397fbf7f533cbad978d649caebf2083a248f7"}]},"branch":"refs/heads/master"},"45dc514ee9e25ddd45acfaa074459b7cc8cbd0f7":{"kind":"REWORK","_number":5,"created":"2022-11-22 03:59:09.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/40/850840/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/850840/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/850840/5"}}},"commit":{"parents":[{"commit":"f8fa57a8b974f30cdda8367ebd470279871ea136","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8fa57a8b974f30cdda8367ebd470279871ea136"}]}],"author":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-25 08:03:55.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2022-11-22 04:03:50.000000000","tz":-300},"subject":"Fujitsu Driver: Add volume migration support","message":"Fujitsu Driver: Add volume migration support\n\nAdded support for volume migration in the Fujitsu driver.\n\nAdded below function:\nupdate_migrated_volume\n\nChange-Id: Ifbb1d7d5271ef62cc12f4de8d916661bdba67640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/45dc514ee9e25ddd45acfaa074459b7cc8cbd0f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/45dc514ee9e25ddd45acfaa074459b7cc8cbd0f7"}]},"branch":"refs/heads/master"},"98a16eafd9ffe1a300785d43a14c8becee7604d1":{"kind":"REWORK","_number":6,"created":"2023-03-24 02:39:34.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/40/850840/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/850840/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/850840/6"}}},"commit":{"parents":[{"commit":"f8fa57a8b974f30cdda8367ebd470279871ea136","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8fa57a8b974f30cdda8367ebd470279871ea136"}]}],"author":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-25 08:03:55.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2023-03-24 02:43:35.000000000","tz":-240},"subject":"Fujitsu Driver: Add update migrated volume support","message":"Fujitsu Driver: Add update migrated volume support\n\nAfter migration of a volume to fujitsu eternus backend,\nupdate the back-end name in provider locationto its original\nvolume name.\n\nAdded below function:\nupdate_migrated_volume\n\nChange-Id: Ifbb1d7d5271ef62cc12f4de8d916661bdba67640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/98a16eafd9ffe1a300785d43a14c8becee7604d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/98a16eafd9ffe1a300785d43a14c8becee7604d1"}]},"branch":"refs/heads/master"},"005f5659dccb970a08829a9f364bf1fd16465d0a":{"kind":"REWORK","_number":7,"created":"2023-06-07 08:40:14.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/40/850840/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/850840/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/850840/7"}}},"commit":{"parents":[{"commit":"f8fa57a8b974f30cdda8367ebd470279871ea136","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8fa57a8b974f30cdda8367ebd470279871ea136"}]}],"author":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-25 08:03:55.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2023-06-07 08:44:07.000000000","tz":-240},"subject":"Fujitsu Driver: Add update migrated volume support","message":"Fujitsu Driver: Add update migrated volume support\n\nAfter migration of a volume to fujitsu eternus backend,\nupdate the back-end name in provider location to its original\nvolume name.\n\nChange-Id: Ifbb1d7d5271ef62cc12f4de8d916661bdba67640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/005f5659dccb970a08829a9f364bf1fd16465d0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/005f5659dccb970a08829a9f364bf1fd16465d0a"}]},"branch":"refs/heads/master"},"469a4989dbd6e6183eb7a2b07827ce617010b5ed":{"kind":"REWORK","_number":8,"created":"2023-06-07 09:17:50.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/40/850840/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/40/850840/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/40/850840/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/40/850840/8"}}},"commit":{"parents":[{"commit":"f8fa57a8b974f30cdda8367ebd470279871ea136","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f8fa57a8b974f30cdda8367ebd470279871ea136"}]}],"author":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-25 08:03:55.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2023-06-07 09:21:44.000000000","tz":-240},"subject":"Fujitsu Driver: Add update migrated volume support","message":"Fujitsu Driver: Add update migrated volume support\n\nAfter migration of a volume to fujitsu eternus backend,\nupdate the back-end name in provider location to its original\nvolume name.\n\nChange-Id: Ifbb1d7d5271ef62cc12f4de8d916661bdba67640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/469a4989dbd6e6183eb7a2b07827ce617010b5ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/469a4989dbd6e6183eb7a2b07827ce617010b5ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
