)]}'
{"id":"openstack%2Fcinder~851170","triplet_id":"openstack%2Fcinder~master~I03575e97f9e57685480b62d8b14060262428c8ff","project":"openstack/cinder","branch":"master","topic":"fujitsu-driver-update","attention_set":{},"removed_from_attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2023-04-25 07:23:13.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"30615":{"account":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"last_update":"2023-04-25 07:23:13.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"33609":{"account":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"last_update":"2023-09-15 01:14:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I03575e97f9e57685480b62d8b14060262428c8ff","subject":"Fujitsu Driver: Add manage volume support","status":"ABANDONED","created":"2022-07-27 07:41:59.000000000","updated":"2023-09-15 01:14:33.000000000","total_comment_count":42,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2a672d042ea3cab1b7c54b35b0a1717398b48867","_number":851170,"virtual_id_number":851170,"owner":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-18 05:19:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":1,"date":"2023-04-25 07:23:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},{"value":-1,"date":"2023-09-14 16:49:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-27 07:44:19.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-07-27 07:44:33.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-07-27 08:05:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-27 08:11:21.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-07-27 08:23:18.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-07-27 08:34:21.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-27 08:36:46.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-27 08:59:47.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-07-27 09:27:46.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-27 09:29:07.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-07-27 10:02:24.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-27 10:14:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-28 05:56:09.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-07-29 22:53:55.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-08-05 20:49:05.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-06 04:32:47.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-06 06:25:12.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-10 10:29:18.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-12-10 11:26:57.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-10 11:35:32.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-12-13 22:27:34.000000000","updated_by":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"reviewer":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"state":"CC"},{"updated":"2023-03-22 03:24:53.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2023-04-14 06:20:45.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-14 06:57:11.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-04-14 07:16:10.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-04-14 08:03:51.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-04-14 10:54:50.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-04-14 15:05:53.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2023-04-14 17:26:12.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-17 07:02:40.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2023-05-18 05:26:30.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-05-22 18:10:52.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-09-14 16:49:08.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2023-09-14 21:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"}],"messages":[{"id":"7ea98e7f1034984aaf87765c3bfcd1bfc1049939","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-07-27 07:41:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba90c79e3d266adf854d552a542711105ed41f00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-07-27 07:44:16.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f6da987747c541d2b0cfe34fab2421466d21421d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-27 07:44:19.000000000","message":"Patch Set 1:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":1},{"id":"531e203568afea4b407c7d8b8a7625586c5a47ba","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-27 07:44:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"b8ba91bfc4e51d756e9c54e5ce369f5762608174","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-07-27 07:46:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"8c6fcc5d35b3f3e0328bb72d80b32492b6126010","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 08:05:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f2b7df956ab48f491283e0b33db0035\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/48e4605edb9747d8b8fe76465bba12e7 : SUCCESS in 18m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba4bff412f524683b63d7469bb2d7bf1 : SUCCESS in 19m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"53915aab3d9600d772f30ac8b0e994138edbdc8b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-07-27 08:11:21.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/70/851170/2/check/iscsi-driver-alletra-6k/7d0a390/7d0a390.html : FAILURE in 25m 27s","accounts_in_message":[],"_revision_number":2},{"id":"82f9b083f2de7cda084a559b685c12de95cca0a2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-27 08:23:18.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1658907727113 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"ec9e9fbb06c2d582e8b2d572533ec5b9090fc104","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-07-27 08:34:21.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"47ef97e0b198bf2521c036511f2d7acfebd8f575","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-07-27 08:36:46.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/2/check/fujitsu-eternus-dx-iscsi/7750f4a : FAILURE in 50m 24s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/2/check/fujitsu-eternus-dx-fc/913778e : FAILURE in 51m 53s","accounts_in_message":[],"_revision_number":2},{"id":"539fdb582de372c4ba0b1b6b227c45da6218305e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-07-27 08:59:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/851170/2/iscsi/ : SUCCESS in 1h 14m 58s\n- cinder-smb http://cloudbase-ci.com/cinder/851170/2/smb/ : SUCCESS in 1h 06m 50s","accounts_in_message":[],"_revision_number":2},{"id":"ec1600aa33fb5925c42996b7310badfbf92893c6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-07-27 09:00:10.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1658907867117 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"f9f428c8236173d1ee51d33bf3868b2275104785","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-07-27 09:27:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/851170 : SUCCESS in 1h 38m 29s","accounts_in_message":[],"_revision_number":2},{"id":"c37c77ef31cdcac7fa85541e31535c27ba1207b7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-27 09:29:07.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/70/851170/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/96e9ee5/ : SUCCESS in 1h 28m 01s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/70/851170/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/9fbaec2/ : SUCCESS in 1h 43m 00s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/70/851170/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/e1b2e7b/ : FAILURE in 27m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4e216475948d568cfa6a4c7691cba6c57f904b2e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-07-27 10:02:24.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/10855eb3a8bb4034aef2c3f6cffb5ae1 : POST_FAILURE in 1h 23m 45s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/8f9af038166547c5885014491da31241 : POST_FAILURE in 2h 11m 00s","accounts_in_message":[],"_revision_number":2},{"id":"9bfd1b773b8d39690a904e769bba7bcb3a078e94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 10:14:42.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/36c8db001668443a862d3e2810499db5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec0d4cb40c86483188eb0a35460d9f18 : FAILURE in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e861314195647518a936039611afdb5 : SUCCESS in 13m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f4359644e3841e5b54534cccb988847 : SUCCESS in 15m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1679d9b3f93d4ff18cd47ee8b4ce8bd3 : SUCCESS in 17m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/006eb0a8edf245fb9726264f67f8f3dc : SUCCESS in 13m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/3a4f5d39c5d34371b348dec84a0ff5ba : SUCCESS in 59m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2c411d7cf4284ee0a9f7e047a23bfc9f : SUCCESS in 1h 25m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/713edcc661c5406f8079a1e22429aa6a : SUCCESS in 57m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ea72a122fbd84799adf56ea5bba43703 : SUCCESS in 17m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f097913d7a784173b41a1373b4e6f632 : FAILURE in 5m 44s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a5959287e0a44ab5ae16811a684631d0 : FAILURE in 7m 09s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eeff020f7b744b79a0690982cb1f5020 : SUCCESS in 6m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/47cd13faca894a4484393d5cda4135ee : SUCCESS in 6m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d3c03a8ee312489db8b65c5eefcaa981 : SUCCESS in 39m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9f8deec401034b55bd13a7357afc085a : FAILURE in 6m 41s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f17e12ceb3c4489382e6342aa01c9d46 : SUCCESS in 1h 26m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/58f10065b5a94f709a77689b7987565a : FAILURE in 1h 30m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cdf5b051e35c406a8b7583148d984f84 : SUCCESS in 1h 39m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c7591950bd8343e58556ab0c98d73330 : SUCCESS in 1h 17m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5f698c3060ad4a78b05d3126d01e8642 : SUCCESS in 1h 22m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ff38d24edfa0494499b7d9ec91a1c066 : SUCCESS in 52m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cc8ee6221fc04f4187bd74f893dbe60f : SUCCESS in 1h 25m 12s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e9ebce92bf5e46f5850010a5bda0f04e : SUCCESS in 2h 23m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e0f5f8c3fe54fe293db6e8bb065353e : SUCCESS in 51m 13s","accounts_in_message":[],"_revision_number":2},{"id":"9b931c01cf81028b138ec7e9528ab62604abe468","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-07-27 11:01:30.000000000","message":"Patch Set 2:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/282","accounts_in_message":[],"_revision_number":2},{"id":"4d6e6b109cb18be9ea99f16631a8f11e4d2a33c7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-28 05:56:09.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/70/851170/2/check/DellEMC_Unity_FC/c9841be/DellEMC_Unity_FC/1276 : FAILURE in 15m 18s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/70/851170/2/check/DellEMC_Unity_iSCSI/a1b4a23/DellEMC_Unity_iSCSI/1170 : SUCCESS in 2h 11m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"b9c6932a132e9f31b8aa7e819e39c9b31b129887","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-07-29 22:53:55.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/2/20220730052524/NEC-V-FC : SUCCESS in 2h 06m 15s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/2/20220730052524/NEC-V-ISCSI : SUCCESS in 2h 28m 27s","accounts_in_message":[],"_revision_number":2},{"id":"4aeef5fde0c55ab8664e8b322f5ef798c34aadba","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-08-02 06:56:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"66ad09f0d43c6b9f900f5d7e548207344aa228ea","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-02 10:47:22.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/2/check/fujitsu-eternus-dx-iscsi/03c9feb : FAILURE in 3h 48m 34s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/2/check/fujitsu-eternus-dx-fc/18887f1 : FAILURE in 3h 50m 29s","accounts_in_message":[],"_revision_number":2},{"id":"307e52495b6ba9c8c09b477b4368808cc5db8ebe","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-03 09:44:07.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/2/check/fujitsu-eternus-dx-iscsi/f80b95d : FAILURE in 1h 34m 21s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/2/check/fujitsu-eternus-dx-fc/5ab1a2b : FAILURE in 1h 34m 01s","accounts_in_message":[],"_revision_number":2},{"id":"286862336b4f8984db88c92b9bc5b58d1ac14dbf","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-08-05 20:49:05.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/2/20220806021904/NEC-FC : SUCCESS in 3h 29m 10s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/2/20220806021904/NEC-ISCSI : SUCCESS in 3h 29m 58s","accounts_in_message":[],"_revision_number":2},{"id":"7ee7fcc56e0ea069eb0e9cc45689924e02f700de","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-11-21 03:01:38.000000000","message":"Patch Set 2:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/2/check/fujitsu-eternus-dx-iscsi/fdf6675 : SUCCESS in 1h 51m 33s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/2/check/fujitsu-eternus-dx-fc/f0f5d6b : SUCCESS in 1h 59m 10s","accounts_in_message":[],"_revision_number":2},{"id":"05ea330e2d87d22ad375a194084bc0f58c23ca83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-06 02:51:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9fff68312561063a211592aa60406a9e41acbb7c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-12-06 02:51:54.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1832 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"2bf28a1e3c0b9fbe4ea39016ece781d67ed012a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 03:39:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa37f17e621f4f738a4377f6c56c3dda\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2aa5d03bc8242c0ad6decab7cf95b9e : SUCCESS in 23m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/451ddaf2215c4384befdae848868a324 : SUCCESS in 22m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5b24eb477271c90146ae5d76877c7bc037dcf332","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-12-06 04:32:47.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/70/851170/3/ : SUCCESS in 1h 32m 39s","accounts_in_message":[],"_revision_number":3},{"id":"4dd141a7eb8cbfe12c2ba8baef3a82148aa21916","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-06 04:45:06.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/70/851170/3/check/iscsi-driver-alletra-6k/fb20040/fb20040.html : FAILURE in 1h 50m 42s","accounts_in_message":[],"_revision_number":3},{"id":"c6a1818fb2db5c77c82110459ae8feaa2332cfa8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-12-06 04:45:09.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/70/851170/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/696984c/ : SUCCESS in 1h 52m 19s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/70/851170/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/43b68b8/ : SUCCESS in 1h 40m 13s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/70/851170/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/d5dbd09/ : SUCCESS in 1h 37m 50s","accounts_in_message":[],"_revision_number":3},{"id":"694e907f30bd6c5d8a90927bcd631c0250106548","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 04:55:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84eb273704084880a404a8987e04fa85\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00042503c1b94af590c0b54bf86d51f5 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5beb3fa2d1d0481f8ed05f98b044a668 : SUCCESS in 10m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a25ec770ab59444aabe07795457d70bb : SUCCESS in 10m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91df10a5cc1444d289ba417e2558e928 : SUCCESS in 12m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/69e8fe74c1d04abf97af3702dc6364df : SUCCESS in 55m 33s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4907bac4662a4fcb890eefee539a6a1b : SUCCESS in 56m 00s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5b0b44498fa3410790b1d8ddfaa30c44 : SUCCESS in 1h 20m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/132be494ba1c468eb07760bf29a17fcb : SUCCESS in 54m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/eb76bc0b30d84806b50e1be22a4fea0b : SUCCESS in 14m 50s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3818af54de784116ba5e8a55a1bda5ad : SUCCESS in 5m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f0b875eb2fbf42caaf374f946807a236 : FAILURE in 6m 21s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/04ba85f8e922497983a86a88e185f257 : SUCCESS in 5m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a1c102608c7b44908e0a2801c22d3c2e : SUCCESS in 6m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7648955954dc4929a323185594aa2de4 : SUCCESS in 36m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/49ae88f00e874f058e0a29965d7c99b0 : FAILURE in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a102e8f260fc404383695a51bcb9fddb : SUCCESS in 1h 14m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0ba7bfd9832d4485a482923db69f20fe : FAILURE in 1h 14m 50s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/705e8d718b224dd1b73dfe3fcb4ccb24 : SUCCESS in 1h 33m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0f5b74a65fc3420fb5051887dc4a0d44 : SUCCESS in 1h 03m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e70a13f2362c4ef0b8b7102a57eb56b2 : SUCCESS in 1h 27m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2abe7abc9c824b33ad079e0cada0e186 : SUCCESS in 29m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/81c6e22e45864b1e85bff2c1c493f480 : SUCCESS in 1h 22m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cfe44bfdef7948eda67ef1e7ad1308c6 : SUCCESS in 2h 02m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a93e04831934594b45242e2d076ee22 : SUCCESS in 38m 07s","accounts_in_message":[],"_revision_number":3},{"id":"8e76e0472086ae03b23ed99b53c52b11fe5bae53","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-06 05:08:57.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/70/851170/3/check/DellEMC_Unity_FC/431b0b7/DellEMC_Unity_FC/2065 : SUCCESS in 1h 36m 04s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/70/851170/3/check/DellEMC_Unity_iSCSI/8662e12/DellEMC_Unity_iSCSI/1912 : SUCCESS in 2h 16m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"fada6ea5be5d31451b14305581ca0e0f24491638","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-06 05:10:04.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/70/851170/3/check/iscsi-driver-nimble/fc70658/fc70658.html : FAILURE in 2h 11m 37s","accounts_in_message":[],"_revision_number":3},{"id":"de8b3756fd2b07e39d8cb0bf28882505ac22eb68","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-06 05:34:58.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/851170 : SUCCESS in 2h 40m 19s","accounts_in_message":[],"_revision_number":3},{"id":"aa2bbabab133e9600ec3f87d4c0b30ee467adc40","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-06 05:46:15.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/851170 : FAILURE in 2h 50m 47s","accounts_in_message":[],"_revision_number":3},{"id":"8ae2ebeadc13d8dbb5c1d9dde52a5236bf41b51b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-12-06 06:25:12.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-70-851170-3/ns5_nfs : SUCCESS in 3h 30m 8s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-70-851170-3/ns5_iscsi : SUCCESS in 3h 32m 55s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"a6b5b56616b31f4e541371b2188a6921ca367eef","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-12-06 06:41:20.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/3/check/fujitsu-eternus-dx-iscsi/0bd1090 : FAILURE in 3h 48m 24s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/3/check/fujitsu-eternus-dx-fc/30b524d : FAILURE in 3h 48m 53s","accounts_in_message":[],"_revision_number":3},{"id":"4ecc0017457b21d85156b5f379ccc2ffce90ad7f","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-06 07:25:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a290727ebd174d2c22928b8a8e3b2a5dc1644daa","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-06 07:28:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a4134d6e4aed9987e85b598c13618841bf5f36ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-06 09:37:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f905c461e40fcb2a0bbdcd9039fe3e3757b68dab","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-12-06 09:37:59.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1833 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"ab57a5ac33cf27dc98daa18c5928a4e2043acc9b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-06 09:42:54.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"2507248f5a4f94a0cf83c750933d5696d7369c57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 10:07:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/154fe9cfbbd4461bb75dd3133df64e20\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/31721e6dfe4941e59e57ce7594134a01 : SUCCESS in 25m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/19092b9d996c47e3900ba1998da3a1b7 : SUCCESS in 26m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3b8b92cb3eec3e33b91de76fb88e2fc4d7a671c9","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-12-06 11:17:48.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/70/851170/4/ : SUCCESS in 1h 31m 28s","accounts_in_message":[],"_revision_number":4},{"id":"27cd0c381c65a3f20848cc42715bc0954227d03b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-12-06 11:25:17.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/70/851170/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/e9dcb45/ : SUCCESS in 1h 46m 29s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/70/851170/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/156fe4f/ : SUCCESS in 1h 41m 27s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/70/851170/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/be8a5a4/ : SUCCESS in 1h 46m 19s","accounts_in_message":[],"_revision_number":4},{"id":"7694899c86381b120c29741ce4d7bfbc6f730e3c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-06 11:42:18.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/851170 : SUCCESS in 1h 59m 40s","accounts_in_message":[],"_revision_number":4},{"id":"5d6c065ff131264a4e908950d442ae27ec9d4714","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-06 11:49:24.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/70/851170/4/check/DellEMC_Unity_FC/208f673/DellEMC_Unity_FC/2066 : SUCCESS in 1h 36m 25s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/70/851170/4/check/DellEMC_Unity_iSCSI/2af5141/DellEMC_Unity_iSCSI/1913 : FAILURE in 2h 11m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"a971ad1c7772d96e68db0b8b11b1f7d3c1e21f5c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-06 11:55:08.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/70/851170/4/check/iscsi-driver-alletra-6k/0c5f3c5/0c5f3c5.html : FAILURE in 2h 14m 42s","accounts_in_message":[],"_revision_number":4},{"id":"b19d4a270b35b5b347cbbed169967b8d99640a07","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-06 12:10:32.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/851170 : SUCCESS in 2h 28m 53s","accounts_in_message":[],"_revision_number":4},{"id":"82a2f711365e5ef7de2471daf157e59cbafdf01d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 12:12:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a337510417fe4e90b5eab83a29276e1b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/084704ca511c44b897d2f8dd560b2a15 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be42e6f491d74ea0b084e83bb7afb724 : SUCCESS in 9m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/edbae46925564feda1a5050a1b772f96 : SUCCESS in 10m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f615a279e1ed4c15a908b7239577142d : SUCCESS in 13m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/194747672ba0415e9cfedc53762d5b4a : SUCCESS in 56m 42s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/110d515080544e82ada128e56a9aa98c : SUCCESS in 1h 01m 20s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9cc1e63dd750452cafdf97405305bc9c : SUCCESS in 1h 25m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71313c8d83964880b744a320e8cb9860 : SUCCESS in 57m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dc9f1bc01b194c709b0ec4fb4152b0ed : SUCCESS in 16m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0aa4d57fdbe841738bcb466c5b9e815b : SUCCESS in 5m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f22b691615234f2995e64398ebe69d19 : FAILURE in 6m 30s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/86e9322c15b34b6aa52d60e48cfda63a : SUCCESS in 5m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6c8ded9225994fd79b459e3d9134d331 : SUCCESS in 5m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fd3d3ed8af774777b12ecf90482236b7 : SUCCESS in 1h 00m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/52f43cecca1d4d1b877aab786bb45bd8 : FAILURE in 5m 32s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cb05f601930f468a8bc500c8d6bbf894 : SUCCESS in 1h 24m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5a5f2be01cfa48c284918984d7e664b7 : FAILURE in 1h 46m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b1dda3876cbb4b3496dca23fe222dccf : SUCCESS in 1h 42m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dcc60eba5d114a92a1b198c651e74517 : SUCCESS in 1h 14m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8deabc2d1e3348f0809822702754d15c : SUCCESS in 1h 48m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/40faae2d39b94358a7797813479ae8e7 : SUCCESS in 38m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/78eb1a7cc9c448a694f4fdb3ffba26b0 : SUCCESS in 1h 37m 59s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/07293df6ef2044a4a81ab0db28f3ebf7 : SUCCESS in 2h 26m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6898d7d75f7d435da8cb400244fbf258 : SUCCESS in 50m 34s","accounts_in_message":[],"_revision_number":4},{"id":"d0bdbf90f789e008a6a27fab7a7cb6d78b2523f3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-12-06 13:06:03.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-70-851170-4/ns5_nfs : SUCCESS in 3h 27m 42s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-70-851170-4/ns5_iscsi : SUCCESS in 3h 27m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"b2357a33cc5a4ff6290acfdec55f083355df94a1","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-12-06 13:29:37.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/4/check/fujitsu-eternus-dx-iscsi/18d6c84 : FAILURE in 3h 50m 54s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/4/check/fujitsu-eternus-dx-fc/65fb819 : FAILURE in 3h 51m 12s","accounts_in_message":[],"_revision_number":4},{"id":"91ed96df5ee1953a3a1b311fcd347dd6c3467651","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-06 14:06:46.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/70/851170/4/check/DellEMC_Unity_FC/8b82649/DellEMC_Unity_FC/2067 : SUCCESS in 1h 32m 41s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/70/851170/4/check/DellEMC_Unity_iSCSI/d4816d2/DellEMC_Unity_iSCSI/1914 : SUCCESS in 1h 54m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"3ce2d438d06554bfa60ed7349d3405dfb5847ddb","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-06 15:19:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4dfae463f6225a9cb0f221706263588ab8d39ae9","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-12-06 17:19:16.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/4/check/fujitsu-eternus-dx-iscsi/fa8dee1 : FAILURE in 1h 50m 42s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/4/check/fujitsu-eternus-dx-fc/3114c03 : FAILURE in 1h 58m 54s","accounts_in_message":[],"_revision_number":4},{"id":"abb0fe1292d170984bc2223b68214f9cda77a9e1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-12-06 21:03:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/4/20221207023225/NEC-ISCSI : SUCCESS in 3h 29m 55s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/4/20221207023225/NEC-FC : SUCCESS in 3h 30m 41s","accounts_in_message":[],"_revision_number":4},{"id":"6ff90c506dbea59941f66dc0c3d649992ac0817e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-12-07 00:17:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/4/20221207024628/NEC-V-FC : SUCCESS in 6h 01m 53s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/4/20221207024628/NEC-V-ISCSI : SUCCESS in 6h 31m 12s","accounts_in_message":[],"_revision_number":4},{"id":"5fba2f671601a1dc542e9a0dbdd06384c28181a0","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-07 01:47:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e96913894183e48aff74db35fc6eec1588286311","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-12-07 03:24:31.000000000","message":"Patch Set 4:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/4/check/fujitsu-eternus-dx-iscsi/3c06bc8 : SUCCESS in 1h 36m 05s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/4/check/fujitsu-eternus-dx-fc/dbd148a : SUCCESS in 1h 35m 54s","accounts_in_message":[],"_revision_number":4},{"id":"447a88c63e82d78cd95074f4fb47dea40f3abb6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-07 04:00:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"346183800338e7627327138e044e8d4d8f60a53f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-12-07 04:00:51.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1835 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"5f93e9b288d56c365a0992bf5ab66c21cabe539a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 04:26:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ffbe28cca594f9db790110b3a5eecc5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8b10f2243b24fddb2e1e8670122ce08 : SUCCESS in 23m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3ff311b056634e38aae8e3eca9b56618 : SUCCESS in 19m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"401f26f82235d9c8517d5c85447b905668c39d78","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-07 04:28:41.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/70/851170/5/check/iscsi-driver-alletra-6k/3753a26/3753a26.html : FAILURE in 25m 39s","accounts_in_message":[],"_revision_number":5},{"id":"4045b89f24e6efe3c5794d8cb1338954cf4f6710","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-12-07 05:30:47.000000000","message":"Patch Set 5:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/5/check/fujitsu-eternus-dx-iscsi/94be9dc : FAILURE in 16m 07s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/5/check/fujitsu-eternus-dx-fc/abd6b4d : SUCCESS in 1h 29m 25s","accounts_in_message":[],"_revision_number":5},{"id":"0f750f5abf704257cf45bee62e230deeccdec99c","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-07 05:32:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6e3640c7f1b20b71d89fd487876702d350a452f9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-07 05:45:49.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/70/851170/5/check/iscsi-driver-nimble/345436e/345436e.html : FAILURE in 1h 40m 15s","accounts_in_message":[],"_revision_number":5},{"id":"4fded71cfb536120406d11097a083956bff9905a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-12-07 05:48:23.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/70/851170/5/ : SUCCESS in 1h 39m 04s","accounts_in_message":[],"_revision_number":5},{"id":"18d280dccf180309cf67b7919ab6ab6c25388069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 05:52:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c54fabfc6ec4fd0b8fc38b511c6c091\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f17cd201a544b2ca8382b40097623bc : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb0464d3ee3b427595a66f9137b8b50b : SUCCESS in 8m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ce8d57e3f74455c8d4b40e27a557b3e : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26a83f4228f34400b95547b597f50515 : SUCCESS in 11m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/340f819a52a941bab4f0c995fe1dbc2d : SUCCESS in 42m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/07d73b752a4a48f994132c08d33b7cb0 : SUCCESS in 44m 07s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/06d416113f074c919593dbdd9ee7aeb2 : SUCCESS in 1h 10m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15b5dbb2c29843e28c94e665491fd6a6 : SUCCESS in 47m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a36caebd30a745ee96d331d15aa2628d : SUCCESS in 13m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/79c2dfe050cc4437a1e947cad4055207 : SUCCESS in 4m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/83071c820205437abf7d8511bbd1f2bd : FAILURE in 6m 12s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/29e996c3a9e24ba0bbe86daa9d028632 : SUCCESS in 6m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ffac96e22b834ee885cb1889256792fa : SUCCESS in 6m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/524f259d46c34bab86bd62452fdde1d5 : SUCCESS in 35m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a154b1c9989446cb8c34a3984f26cebc : FAILURE in 5m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/296d6e49c3d6475689ea388b347963cc : SUCCESS in 1h 15m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/be00edee576b4928a661a0ab1765c17e : FAILURE in 1h 07m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c171098b941645358bb9a2e1d83a07f4 : SUCCESS in 1h 44m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5ec18037ad8a49369e33dd4b742a61a1 : SUCCESS in 53m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/aa18f1c14df74146b0b86115e0dfa046 : SUCCESS in 1h 05m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ac547506b0ae4a999882b3ad37e2066d : SUCCESS in 29m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bcc3c933e7f541dc8ab583a0901017ae : SUCCESS in 1h 15m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4a94fecc1b474cfeaa633f0f2e36939d : SUCCESS in 1h 37m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56e04b5e28af4d3692cd9d8101cc95c4 : SUCCESS in 47m 44s","accounts_in_message":[],"_revision_number":5},{"id":"2a89f932f1450d0352bb71a64c3cc4324d312821","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-12-07 05:57:27.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/70/851170/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f39351f/ : SUCCESS in 1h 31m 20s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/70/851170/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/839b6dd/ : SUCCESS in 1h 35m 29s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/70/851170/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/414d6e5/ : SUCCESS in 1h 55m 29s","accounts_in_message":[],"_revision_number":5},{"id":"cb8e0afeafa9fa06a881f7648802d273866a804f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-07 06:46:00.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/851170 : SUCCESS in 2h 42m 29s","accounts_in_message":[],"_revision_number":5},{"id":"21032ca1622f9b291b74b9a6890edc94a6c502bc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-07 06:52:02.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/851170 : FAILURE in 2h 49m 50s","accounts_in_message":[],"_revision_number":5},{"id":"0c9c8aa05e1df72557cc93c42528570f0a1fe706","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-12-07 07:14:44.000000000","message":"Patch Set 5:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/5/check/fujitsu-eternus-dx-iscsi/23d3f1c : SUCCESS in 1h 41m 37s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/5/check/fujitsu-eternus-dx-fc/6d1ef9c : SUCCESS in 1h 31m 26s","accounts_in_message":[],"_revision_number":5},{"id":"d7f2f28501dfa700740d160be354ca5d7320296b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-12-07 07:34:21.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-70-851170-5/ns5_nfs : SUCCESS in 3h 31m 17s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-70-851170-5/ns5_iscsi : SUCCESS in 3h 33m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"78decb6ba0715a104c70c9c719ec8913052d6f70","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-12-07 07:39:24.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/5/20221207130046/NEC-FC : SUCCESS in 3h 34m 10s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/5/20221207130046/NEC-ISCSI : SUCCESS in 3h 38m 34s","accounts_in_message":[],"_revision_number":5},{"id":"d96e6b3ab1e46b59014d4138e2dad2f97ca74a51","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-12-07 13:04:34.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/5/20221207154021/NEC-V-ISCSI : SUCCESS in 2h 49m 31s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/5/20221207154021/NEC-V-FC : SUCCESS in 6h 24m 10s","accounts_in_message":[],"_revision_number":5},{"id":"e7740ebd260205cc4d33dafa6c4ea7918b4d4996","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-10 08:21:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e544eb201c3464410a5b07094b1c8c62238ddfc6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-12-10 08:21:24.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1847 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"ce352ce74d8d445011fcb7162f37d7f2d5fe4273","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-10 08:52:17.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b266d0ce71ef4f8783688fa5e894cc20\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46cd89923be54971ad098454ecfeef37 : FAILURE in 29m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2650502967004480b538cb16ed939443 : FAILURE in 27m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1034d1d39bea36185cd8ba35d66f397c29cfe8ca","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-12-10 09:17:40.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/6/check/fujitsu-eternus-dx-iscsi/0f97b8e : FAILURE in 54m 58s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/6/check/fujitsu-eternus-dx-fc/3b2b469 : FAILURE in 55m 46s","accounts_in_message":[],"_revision_number":6},{"id":"b87badeaeb2b2979da9bdcd0829de99e57194dc9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-10 10:03:28.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/70/851170/6/check/iscsi-driver-nimble/dfd5b96/dfd5b96.html : FAILURE in 1h 39m 08s","accounts_in_message":[],"_revision_number":6},{"id":"a69d95103f7c686019f0b88d1ad0ede3a9a96f6c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-10 10:05:14.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/70/851170/6/check/iscsi-driver-alletra-6k/21a3255/21a3255.html : FAILURE in 1h 42m 58s","accounts_in_message":[],"_revision_number":6},{"id":"fb9e610d152aac3431308fae0b7367873b90afd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-10 10:19:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39a0cc53e82d424688de0b5dfb8874aa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25d0b370600b45589b71603c8a49e61f : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f194ee1d669a49ba827148ed16ec9a75 : SUCCESS in 8m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93448694de3d4c29adc41d355bdd98a6 : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdf0279b90ae4d5dbc54f18c5fd07953 : SUCCESS in 13m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/d3014a2165044e779298968fea94e86f : SUCCESS in 1h 05m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/59fbe8278db346f2b4656ece264ae944 : SUCCESS in 1h 04m 58s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6e0c5d8258e74737976987b4c1daa56d : SUCCESS in 1h 20m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e1edda3f5564442b665462ebb21addf : SUCCESS in 48m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f575d9f2862048a29d1bd2f048ede5e8 : SUCCESS in 15m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2219599662ae43b1ba2b86976d91db12 : SUCCESS in 5m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/64cc48a5d7024d30a508eaad1f04603a : FAILURE in 8m 38s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/62ebad17da21468cb7075425ce409907 : SUCCESS in 5m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3d12b2f87d9d408e94b432f790efabd8 : SUCCESS in 6m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ce54319f0c1347bf869da5e108158967 : SUCCESS in 48m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9b3f2921212a434abb8ad3743bd100dd : FAILURE in 5m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8c9573989f9e46a68b98fb2c1e41bcc1 : SUCCESS in 1h 03m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b572ecc789d54680b1e032f7fd002d20 : FAILURE in 1h 20m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/be29c6bdbf884d36bf779653a34e58b2 : SUCCESS in 1h 38m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9d00d8afd19d41e1a741fcebcf54e4e1 : SUCCESS in 1h 01m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/66df947d6abb42f1a531f553a440565c : SUCCESS in 1h 18m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/094c2d162f2c42989d9156994c1ef843 : SUCCESS in 34m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4c1ebaaf4d504ab0a911a3e4f9125cdf : SUCCESS in 1h 17m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8512857c539948a7909e9705e73eacb7 : SUCCESS in 1h 54m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a18088f91d14cca951afb2473f024d5 : SUCCESS in 34m 59s","accounts_in_message":[],"_revision_number":6},{"id":"8de84adb61a652ff9c104fa03bfc7cba4f129299","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-12-10 10:20:52.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/70/851170/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/0547ca4/ : SUCCESS in 1h 29m 31s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/70/851170/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0d32dc5/ : SUCCESS in 1h 58m 19s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/70/851170/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/acfdf79/ : SUCCESS in 1h 34m 54s","accounts_in_message":[],"_revision_number":6},{"id":"1a1f8ff5c13ba6909577523a3778e2a682155952","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-12-10 10:21:50.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/70/851170/6/ : SUCCESS in 1h 51m 50s","accounts_in_message":[],"_revision_number":6},{"id":"22c6d0de1041b05dc0189e075029abc1e4b0ae92","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-10 10:22:03.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/851170 : SUCCESS in 1h 57m 14s","accounts_in_message":[],"_revision_number":6},{"id":"d901182f1a7de31d0afc7f53163d5172abe11a84","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-10 10:24:48.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/851170 : SUCCESS in 2h 00m 17s","accounts_in_message":[],"_revision_number":6},{"id":"00e699c7185c3985c518a0f154c7925c9b8c5f35","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-12-10 10:29:18.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/851170/6/upstream-check/cinder-tempest-plugin-ontap-fcp/4620bb1/ : RETRY_LIMIT in 7m 30s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/851170/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/8567897/ : RETRY_LIMIT in 7m 24s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/851170/6/upstream-check/cinder-tempest-plugin-ontap-nfs/8840c1e/ : RETRY_LIMIT in 7m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"81ba63b1e46e133a59e8c0fe8b2515d6eb3ff4f9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-10 10:31:24.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/70/851170/6/check/DellEMC_Unity_FC/34c9c9e/DellEMC_Unity_FC/2080 : SUCCESS in 1h 42m 47s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/70/851170/6/check/DellEMC_Unity_iSCSI/3ed8549/DellEMC_Unity_iSCSI/1925 : SUCCESS in 2h 09m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"6e642bfa636e6ed768d4156161598828ae31b79a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-12-10 11:26:57.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/70/851170/6/check/DellEMC_PowerFlex/ac7d9cf/DellEMC_PowerFlex/1479 : SUCCESS in 1h 06m 56s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"401c36cb76e833a9ba2e953935cc612e44695e76","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-12-10 11:35:32.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/70/851170/6/check/DellEMC_SC_FC/a1ec5fb/DellEMC_SC_FC/1108 : SUCCESS in 1h 07m 33s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/70/851170/6/check/DellEMC_SC_iSCSI/1120cb9/DellEMC_SC_iSCSI/1092 : SUCCESS in 1h 15m 35s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":6},{"id":"abd97fddaca308d8c49993ec72684204b4a58fe0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-12-10 11:53:41.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-70-851170-6/ns5_nfs : SUCCESS in 3h 29m 38s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-70-851170-6/ns5_iscsi : SUCCESS in 3h 31m 53s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"621566dd2b22608f7fbf026c3b291e903deb7c0e","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-10 12:50:30.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b717777ce9e6e3e27a8168233fe85ab440eda5be","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-10 12:52:29.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b1d178ded3ff46a41c41550e994dbb31c7680c6b","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-12-10 13:46:12.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/6/check/fujitsu-eternus-dx-iscsi/effbf36 : FAILURE in 52m 54s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/6/check/fujitsu-eternus-dx-fc/8231d75 : FAILURE in 53m 09s","accounts_in_message":[],"_revision_number":6},{"id":"f52103ce7bffa8671b728f8b4b3833ae465d7788","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-12 01:27:17.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4018bbf3ec4948ed6e8c027b5c64637562d0468b","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-12-12 02:23:20.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/6/check/fujitsu-eternus-dx-iscsi/d7f732a : FAILURE in 52m 55s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/6/check/fujitsu-eternus-dx-fc/d7037d4 : FAILURE in 55m 28s","accounts_in_message":[],"_revision_number":6},{"id":"a788ac10ede77d7708e24d560fc7ea09a692a719","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-12 02:34:20.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2d04ba09e3b58ff6a146445be23a521b673f78a8","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-12 02:34:44.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a0a1e05db1444c557344ab63d924106d791f28b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 03:24:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cce8ec02c64487bb95202a285aa43a9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8e9f55e52c54fe4b583c256e9706a54 : SUCCESS in 26m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8ec02c5d54604508ac2247547e759f05 : SUCCESS in 23m 58s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"36ea925b15f0ed680a6fb1e0405291c3ecc11519","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-12-12 03:29:18.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/6/check/fujitsu-eternus-dx-iscsi/78ccc52 : FAILURE in 53m 59s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/6/check/fujitsu-eternus-dx-fc/09450a8 : FAILURE in 52m 23s","accounts_in_message":[],"_revision_number":6},{"id":"3cc5f0a8e78cb711551c6359a725a7929dc837de","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-12 03:34:34.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7e2258bb5b0596d3ebd7416ecaa77790f2bf61e1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-12-12 03:36:49.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/851170/6/upstream-check/cinder-tempest-plugin-ontap-fcp/ebc9cba/ : RETRY_LIMIT in 11m 14s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/851170/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/d84082e/ : RETRY_LIMIT in 7m 23s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/851170/6/upstream-check/cinder-tempest-plugin-ontap-nfs/80f1933/ : RETRY_LIMIT in 7m 26s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"20404865deafb25a75565f9cce9057ac168e193d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-12-12 04:26:55.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/6/check/fujitsu-eternus-dx-iscsi/2934d58 : FAILURE in 51m 47s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/6/check/fujitsu-eternus-dx-fc/6ef7240 : FAILURE in 51m 37s","accounts_in_message":[],"_revision_number":6},{"id":"7d4345de48c59f31a5d22b087cb7261310d88e1a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-12-12 04:39:31.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/70/851170/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/77dad8c/ : SUCCESS in 1h 38m 31s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/70/851170/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/8e26ee1/ : SUCCESS in 1h 32m 23s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/70/851170/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/bee7633/ : SUCCESS in 2h 03m 37s","accounts_in_message":[],"_revision_number":6},{"id":"25cea56eb9e961457b1f4d1890da81fe2db8ac75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 04:40:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed93dfdcaad9469bb6af03083ad761f3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a93b8c1efbd4524b0e8f58e230a73f8 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96735f9dbd50468ca4ae09f09f46424c : SUCCESS in 10m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b261bc8f79934810b9d4b30d5607598f : SUCCESS in 10m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/287fee97cc05422ca8b5dad0d7f2a8a9 : SUCCESS in 12m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/c50173550d024139ae841391b1f382d8 : SUCCESS in 1h 01m 09s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/933b6872014e468a8d79325bfbd2b9cc : SUCCESS in 1h 01m 37s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/91f45c7f823d46feb4f9d3b100b9d129 : SUCCESS in 1h 49m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c5436fb8b254c4a828e22c4bebbd819 : SUCCESS in 1h 10m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/448ff8e6e6e7476ca9cb9add21e8a7a7 : SUCCESS in 17m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4a71a0fff08c49859e2c040c979f384c : SUCCESS in 4m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a18337c1849b4fc18af6561a34b5232c : FAILURE in 7m 23s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dde6b0312f8f4788913242b3ad46c391 : SUCCESS in 5m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b53d44b037854101a3486fd293ec67eb : SUCCESS in 6m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/91ebb35cba224090a7fe14366288bf3b : SUCCESS in 53m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ae9d5f9a2c404b04aae803c9a6399047 : FAILURE in 6m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5157f16e4e4f4d318c582731144a166b : SUCCESS in 1h 28m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/347d326125be4b789146f4f3c2becc7b : FAILURE in 1h 54m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/25d219ed6d984944a09d2759f7994e38 : SUCCESS in 1h 50m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c35535e333a14c90b4abc563ac360548 : SUCCESS in 1h 05m 30s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ea8079d196064e83a776223a55642732 : SUCCESS in 1h 30m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9d6064f646424015ad2cf80a5583d239 : SUCCESS in 35m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bc69a9d53a714ae6a094bb9f70e331f3 : SUCCESS in 1h 28m 30s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3ded3aff56f046c6b535e043d1b8ffc6 : SUCCESS in 1h 45m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a4987a253f84d559957d9cccc2e8ee6 : SUCCESS in 32m 57s","accounts_in_message":[],"_revision_number":6},{"id":"b79fc5549658bf9393c18985a7945f5b2dd3fbbf","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-12-12 04:53:47.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/851170/6/upstream-check/cinder-tempest-plugin-ontap-fcp/d1e1287/ : RETRY_LIMIT in 7m 29s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/851170/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/a10bc9d/ : RETRY_LIMIT in 7m 16s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/851170/6/upstream-check/cinder-tempest-plugin-ontap-nfs/49b781a/ : RETRY_LIMIT in 7m 13s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"4302d070329ac298bcf6a3577ea24e879110f9fc","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-12 05:19:29.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4fdd19151d37c9146a72ec342fa3eaa246275774","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-12-12 06:14:35.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/6/check/fujitsu-eternus-dx-iscsi/b74f68a : FAILURE in 52m 51s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/6/check/fujitsu-eternus-dx-fc/673ee7d : FAILURE in 54m 32s","accounts_in_message":[],"_revision_number":6},{"id":"8f3ec67d95fb1f3da61a3ccfd603e21edfcb02b7","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-12 06:45:44.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3c523065ea4f02960ff9a8d35cf73ef24ce70c4e","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-12-12 06:49:16.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/6/check/fujitsu-eternus-dx-iscsi/a4ab5be : FAILURE in 2m 36s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/6/check/fujitsu-eternus-dx-fc/56e7964 : FAILURE in 2m 56s","accounts_in_message":[],"_revision_number":6},{"id":"214418a059c784dc96eee7d45a7ac071070e6b72","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-12 06:56:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"83e61ccb8c00c745445c5c7e97a9c6b1564ca5a1","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-12-12 06:59:42.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/6/check/fujitsu-eternus-dx-iscsi/5839fbc : FAILURE in 2m 31s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/6/check/fujitsu-eternus-dx-fc/71dcb57 : FAILURE in 1m 32s","accounts_in_message":[],"_revision_number":6},{"id":"fe816afe0ea3c27d093f3a3a755bbe39547fd214","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-12 07:06:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a77325552ffa2bdf00ad67b9f0cf441be512bba1","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-12-12 08:39:34.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/6/check/fujitsu-eternus-dx-iscsi/04e8abe : FAILURE in 1h 32m 50s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/6/check/fujitsu-eternus-dx-fc/6d03827 : FAILURE in 1m 56s","accounts_in_message":[],"_revision_number":6},{"id":"57142edeafef1f839b6ef0da9ddd300db7e48929","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-12 08:47:49.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1d5cc55779ccad239e4f881a5d02ce4236cca880","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-12-12 08:47:59.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1849 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"f7b46e85148217bf3001ec87fc5541882b6b7bd6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-12 08:49:32.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"37faaaedcede1d1bb5825a2e992f4a4c89733afc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 09:10:54.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ada5dec83c5b44daade20c28d97de061\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1318e019518540f492f41fcbdcbc9290 : SUCCESS in 20m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9013f33712fb4c4ba7c66809748e2b10 : SUCCESS in 19m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c18ca87eb3c99c0ef2bc67cfaa32f3328c050749","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-12 09:14:30.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/70/851170/7/check/iscsi-driver-alletra-6k/2a03032/2a03032.html : FAILURE in 25m 40s","accounts_in_message":[],"_revision_number":7},{"id":"a330413a449ab1ac06b0624096272ca6cf43bb23","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-12-12 09:44:20.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/7/check/fujitsu-eternus-dx-iscsi/7922736 : ABORTED in 55m 52s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/7/check/fujitsu-eternus-dx-fc/96df06e : FAILURE in 2m 54s","accounts_in_message":[],"_revision_number":7},{"id":"1462d80d2e0e004f6b3ed8a57974f990cf092049","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-12 09:47:28.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ef362d987a1cf433d553aeb668ce6a1b09e879fa","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-12 10:17:36.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/851170 : SUCCESS in 1h 27m 24s","accounts_in_message":[],"_revision_number":7},{"id":"c78219b82ca655e5a90adad42ea89062e3febc76","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-12 10:27:03.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/851170 : SUCCESS in 1h 36m 51s","accounts_in_message":[],"_revision_number":7},{"id":"2cd3a67ddcab534f5ff393964f953722339e2989","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-12-12 10:30:54.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/70/851170/7/ : SUCCESS in 1h 34m 25s","accounts_in_message":[],"_revision_number":7},{"id":"c67063d5383ad2694ab5525c4ec7126553771a04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 10:43:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e333b53915de4d70967e62aea22a134f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a02bbc94a69f48f7bdbd8dd3dd7e6096 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ef59690248d416cb4569e1cffacbb2f : SUCCESS in 10m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e7a99deb5d2440ab75f48d169b7dc10 : SUCCESS in 8m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5378943524e84285887ffb0266af3a03 : SUCCESS in 12m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/c7175456a04f4e41bbfdbe230d7d6066 : SUCCESS in 43m 00s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e9f4f1905d3147088286764fb19910f5 : SUCCESS in 1h 12m 22s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3ee994334ef24c56b03b259d9f0c1fab : SUCCESS in 1h 24m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a739c1eb5b974ae9afd76d5c6815da57 : SUCCESS in 47m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/17e971ebfdef4ac7845ec29f86ef5124 : SUCCESS in 15m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3eece2b215434485bbc386123638a502 : SUCCESS in 4m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/04564c610ebf4ede93ac5413c688383c : FAILURE in 6m 29s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b7395ca6e21a437eb56e9a04a8057c04 : SUCCESS in 5m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/451771c2dcb446dd9f31bb04ff430219 : SUCCESS in 5m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c9305e1725e341dd9ce0d64b28ba7947 : SUCCESS in 41m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fa131cbc20324b33912220e71cd35ad0 : FAILURE in 5m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2c5a9ec50a164ef4b819489e98981852 : SUCCESS in 1h 15m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ce112d6c0242483db79dde95b81b0f2f : FAILURE in 1h 22m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/834f30b22e0344e69ace5b039119bea0 : SUCCESS in 1h 35m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/38c6887c3b23444c8679ce0c5f6e22fd : SUCCESS in 1h 02m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a36e06ec43a04e6ba4d3bc004dddf2c0 : SUCCESS in 1h 27m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a4360b1466c947eab864c61d7ef25633 : SUCCESS in 34m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/77749c02cea84db9b2b944489c213524 : SUCCESS in 1h 27m 35s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/acc425ec831245bb803813a2e98e19ff : SUCCESS in 1h 54m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ec05b2657944e2bbaadd50dc877c2d9 : SUCCESS in 48m 27s","accounts_in_message":[],"_revision_number":7},{"id":"cf52f3ca3c43e083352567a283c7d5c7d9ee3508","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-12-12 10:53:42.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/851170/7/upstream-check/cinder-tempest-plugin-ontap-fcp/5925a1d/ : RETRY_LIMIT in 7m 17s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/851170/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/1aca726/ : RETRY_LIMIT in 8m 06s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/851170/7/upstream-check/cinder-tempest-plugin-ontap-nfs/721be64/ : RETRY_LIMIT in 8m 04s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"de08f0f5ee4f92e7196c6384bb99eef9aa48b738","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-12-12 11:02:14.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/70/851170/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d0a605a/ : SUCCESS in 1h 47m 54s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/70/851170/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/78b20ba/ : SUCCESS in 1h 50m 35s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/70/851170/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/128e6e0/ : FAILURE in 2h 13m 12s","accounts_in_message":[],"_revision_number":7},{"id":"b8b6127633c7973c43a90cdc4fd98d11c969577c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-12-12 11:59:16.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/7/check/fujitsu-eternus-dx-iscsi/3a24121 : FAILURE in 2h 11m 12s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/7/check/fujitsu-eternus-dx-fc/7461d42 : FAILURE in 2h 09m 04s","accounts_in_message":[],"_revision_number":7},{"id":"99f1e0e6ac8c5dced4038f50f076fcb5f19e9220","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-12-12 12:21:40.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-70-851170-7/ns5_nfs : SUCCESS in 3h 30m 58s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-70-851170-7/ns5_iscsi : SUCCESS in 3h 33m 19s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"30f3f0b9d33d078d23cbfe2385a8ea5fb1e7dc1a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-12 17:53:58.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/70/851170/7/check/DellEMC_Unity_FC/9a89b0a/DellEMC_Unity_FC/2082 : SUCCESS in 1h 37m 29s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/70/851170/7/check/DellEMC_Unity_iSCSI/23a5ce5/DellEMC_Unity_iSCSI/1931 : FAILURE in 12m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"23c2b803e1f7ecbc86113c478da98d3801cbb223","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2022-12-13 22:27:34.000000000","message":"Patch Set 6:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/70/851170/6/fc : FAILURE in h m 0s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/70/851170/6/iscsi : FAILURE in h m 0s","accounts_in_message":[],"_revision_number":6},{"id":"49b4c3b6978d9973256f10812d9481c780ef50d7","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2022-12-13 22:27:36.000000000","message":"Patch Set 7:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/70/851170/7/fc : FAILURE in h m 0s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/70/851170/7/iscsi : FAILURE in h m 0s","accounts_in_message":[],"_revision_number":7},{"id":"74d12939ce0a875211bddfb4d2f4c22a591e9470","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-12-15 07:00:43.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"bb6d5640496663679f465be552cbe690f8d6acb8","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-12-16 10:46:08.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/7/20221216130047/NEC-ISCSI : SUCCESS in 3h 28m 01s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/7/20221216130047/NEC-FC : SUCCESS in 6h 45m 18s","accounts_in_message":[],"_revision_number":7},{"id":"3650b02a78d1f5ad1eefdd2807ec2c2464be569e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-12-16 14:48:22.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/7/20221216173645/NEC-V-ISCSI : SUCCESS in 3h 04m 54s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/7/20221216173645/NEC-V-FC : SUCCESS in 6h 11m 33s","accounts_in_message":[],"_revision_number":7},{"id":"56c1fbca789d33f18a2651eca0be7b2d378d17ab","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-03-22 03:24:53.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"be27ae4d65ef8efe65b15ac1d99343cc30e611bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-04-14 05:22:48.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":8},{"id":"4fae50d7d6604563703d97b95fbbf47af1962c26","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-04-14 05:35:08.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"364428fbb70992275b3824717d4b03b2e29b23a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 05:36:35.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d48dfd30c9b14830bfbd03648f9ede48\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/526daf8164e748d99cfd57f10bac9c46 : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bee4e107d745474b9021e458e30c3de0 : SUCCESS in 12m 30s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9959784ebd431e45371ebdb06e604c48f1f05364","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-04-14 06:20:45.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/adfd185cbd1c451db50d62d26f88210a : SUCCESS in 56m 29s","accounts_in_message":[],"_revision_number":8},{"id":"9cde4da729f67f5f9d21a812b2982e504bf7675f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-04-14 06:30:07.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/70/851170/8/check/dsvm-tempest-cisco-zonemanager-job/8a908b6/ : FAILURE in 1h 05m 05s","accounts_in_message":[],"_revision_number":8},{"id":"e5c87c5fbf5f338f12a19d523c0203a3d524fe0b","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-04-14 06:52:27.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/70/851170/8/ : FAILURE in 1h 28m 39s","accounts_in_message":[],"_revision_number":8},{"id":"e9987a7625d70662ba44459ad91935f36128e60d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-04-14 06:56:02.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/851170 : SUCCESS in 1h 30m 54s","accounts_in_message":[],"_revision_number":8},{"id":"efcd040a39747b78d6906fd42474db498eb23b2a","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-04-14 06:57:11.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/73301ab989754501b62f38718e3e248a : SUCCESS in 1h 23m 11s","accounts_in_message":[],"_revision_number":8},{"id":"293768aeea3519de46b57613faf6bfc144827fa9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-14 07:05:29.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/70/851170/8/check/iscsi-driver-alletra-6k/74f6a2b/74f6a2b.html : FAILURE in 1h 41m 31s","accounts_in_message":[],"_revision_number":8},{"id":"b14ed67b14dbdb100f2d37a16b602c7446bfb0e8","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-04-14 07:12:06.000000000","message":"Patch Set 8:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/8/check/fujitsu-eternus-dx-iscsi/d4a3593 : SUCCESS in 1h 38m 46s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/8/check/fujitsu-eternus-dx-fc/f10d75c : SUCCESS in 1h 48m 40s","accounts_in_message":[],"_revision_number":8},{"id":"3effbc4549b06ffd1cb235c7bb66f99f4ae58830","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 07:14:19.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0832f7f6879d4b589c98fa22ceb17733\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41c75aba33a8454484db4708c23e8b79 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93572926a6d84e3b9321b1edd631e9fb : SUCCESS in 12m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3da4532b6d4845a4a22cf54490bcd5a7 : SUCCESS in 10m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82c15efb363144b2aff433b43be735ca : SUCCESS in 12m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/d74acf1b61704783a9d6e6b792f1d4e9 : SUCCESS in 59m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6b1b559a87c34e3cb874fc47f6b241b1 : SUCCESS in 1h 27m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06d5a2c9d07142f99f00b683a4ee32cb : SUCCESS in 59m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2efdeec832d4c65bb87f55a59aaec12 : SUCCESS in 9m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ea0f97f0b44d46a193b558d081be2c92 : SUCCESS in 17m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f427f54960424983856850e99b0c2b03 : SUCCESS in 4m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bedfced1bdbe4521a618eeb81271b47b : FAILURE in 6m 50s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/59a92b7cce7f41bf9d0fb38ae83f07b4 : SUCCESS in 6m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/25c688628c404714a885b17b9996771b : SUCCESS in 5m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6ee06e896b3e4d08b5b7ea2a3533040b : SUCCESS in 36m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7f38e8b0853846c6b2d6e064f98dd99a : SUCCESS in 6m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e1eeabaac1274cf8a736a590dfe64e61 : SUCCESS in 1h 04m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/47eb8df7161c4bec86c9eede6f3937fc : FAILURE in 1h 29m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f05a338ac2824c95b5bc572e59ac69d4 : SUCCESS in 1h 35m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/15785973976040eab9b3441542f39830 : SUCCESS in 1h 10m 39s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e650508bdd114d91b46004b90ee086b4 : FAILURE in 1h 28m 45s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4fc684227e4840fcae7e8343b5b9e141 : FAILURE in 28m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/57428aef5e2146cdbf85e4e2eeae766c : SUCCESS in 1h 07m 41s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/568a2178ae284ea0947a8d127b371f98 : SUCCESS in 1h 47m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49fcd3254aae4ec48711e1ebdb27f1cb : SUCCESS in 57m 31s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26166059b46f42078b1776b3121e417d : SUCCESS in 56m 16s","accounts_in_message":[],"_revision_number":8},{"id":"9d666019161112d9959c924a86074678c7ac5bd8","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-04-14 07:16:10.000000000","message":"Patch Set 8:\n\nBuild failed.","accounts_in_message":[],"_revision_number":8},{"id":"338234e4adcafc19e06ed6be76421d0eb9af3ceb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-04-14 07:27:14.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/70/851170/8/check/DellEMC_Unity_FC/5e5944e/DellEMC_Unity_FC/12 : SUCCESS in 1h 34m 25s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/70/851170/8/check/DellEMC_Unity_iSCSI/f298156/DellEMC_Unity_iSCSI/12 : SUCCESS in 1h 53m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"63ba3a72704e7a3780d1d11a7d0c5e653affb41e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-04-14 07:28:09.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/70/851170/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/c5a8f30/ : SUCCESS in 1h 51m 15s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/70/851170/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/f8d0dd7/ : FAILURE in 2h 04m 06s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/70/851170/8/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/d844555/ : SUCCESS in 1h 51m 04s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/70/851170/8/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/ba06697/ : FAILURE in 29m 24s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"141dc4205d0f3ad87aa2ea2ec6798fa4015789e4","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-04-14 08:03:51.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-851170-8/FC/ : SUCCESS in 2h 32m 54s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-851170-8/iSCSI/ : SUCCESS in 1h 46m 23s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"7801c07f401f99f131959974deb6b8ca184990e1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-04-14 08:21:38.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/851170/8/upstream-check/cinder-tempest-plugin-ontap-fcp/69239c0/ : SUCCESS in 1h 00m 13s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/851170/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/ce8eee4/ : SUCCESS in 1h 06m 34s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/851170/8/upstream-check/cinder-tempest-plugin-ontap-nfs/2bd9b1a/ : SUCCESS in 44m 02s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/851170/8/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/1878f0b/ : FAILURE in 44m 57s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"c569015c9d6c4475d7acaa053386275aeb3e1893","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-04-14 08:58:26.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/70/851170/8/check/DellEMC_PowerFlex/13bd176/DellEMC_PowerFlex/13 : SUCCESS in 1h 43m 46s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"636f276f6905f8e66e6f0f5f60a22e0ad004d7b4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-04-14 09:04:58.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-70-851170-8/ns5_nfs : SUCCESS in 3h 22m 59s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-70-851170-8/ns5_iscsi : SUCCESS in 3h 26m 23s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"9621e90a58b38db85e0eb33b3c43e4e624513a12","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-04-14 10:54:50.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/851170-8/HPMSA-iSCSI/ : SUCCESS in 1:35:51\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/851170-8/HPMSA-FC/ : SUCCESS in 1:16:13","accounts_in_message":[],"_revision_number":8},{"id":"5413f38a7a16b53c758ec5c0769131d961bede88","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-04-14 15:05:53.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/851170-8/Seagate-iSCSI/ : FAILURE in 5:23:42\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/851170-8/Seagate-FC/ : SUCCESS in 1:37:53","accounts_in_message":[],"_revision_number":8},{"id":"64932dd9522ff1e4e857fd91edeb3dfe4821ab9a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-04-14 17:26:12.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/70/851170/8/check/DellEMC_PowerMAX_FC/2681076/DellEMC_PowerMAX_FC/9 : FAILURE in 9h 23m 11s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/70/851170/8/check/DellEMC_PowerMAX_iSCSI/62a0e50/DellEMC_PowerMAX_iSCSI/9 : FAILURE in 10h 11m 31s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"6c0c47658d2e4acb50a76ebea3000cc9e7ca24e9","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-04-14 20:47:42.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/8/20230415023346/NEC-V-ISCSI : SUCCESS in 2h 36m 57s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/8/20230415023346/NEC-V-FC : SUCCESS in 3h 13m 53s","accounts_in_message":[],"_revision_number":8},{"id":"a576eb1b6a28ffdccd0dd0f2d69b54f4410e0623","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-04-15 21:23:05.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/8/20230415230712/NEC-ISCSI : SUCCESS in 3h 18m 35s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/8/20230415230712/NEC-FC : SUCCESS in 7h 15m 50s","accounts_in_message":[],"_revision_number":8},{"id":"d38ea9abaace36d58c8a721195e9edcadf2a380d","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-04-17 07:02:40.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"9730fb07d5490589bf70faad9ff42636fef724f0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-18 14:26:52.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/70/851170/8/check/iscsi-driver-nimble/3b02dc1/3b02dc1.html : FAILURE in 1h 43m 08s","accounts_in_message":[],"_revision_number":8},{"id":"09b47304d4524f5b70fd9a7401ac86c9b8dd0b61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-04-23 03:15:35.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"40feca07e13e5dc9fdd9b99c8c59b6925fb202a1","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-04-23 03:29:28.000000000","message":"Patch Set 9:\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"80f10c1150ec1e092c5a7c0547c1049b092cb311","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-04-23 03:34:46.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-70-851170-9/ns5_nfs : FAILURE in 18m 38s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-70-851170-9/ns5_iscsi : FAILURE in 18m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"61d339f1ecb0aaf18669508bfd4ab6959bb8c584","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-04-23 03:36:49.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/fce751a5241c42459abc9ec2e5ec8f25 : POST_FAILURE in 20m 31s","accounts_in_message":[],"_revision_number":9},{"id":"5006126145258f7e79a3f557ed1c46c7aba33c72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-23 03:43:49.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3efca16081f944ce9197fca9b65a52a8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0d4cfe2e7c14074a8a25f880483c020 : SUCCESS in 12m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/084e276ee984409d838248c7110bdc52 : SUCCESS in 25m 37s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ab519095ddf522f1bed7699592af3bf351314c3d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-04-23 03:45:30.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/9/check/fujitsu-eternus-dx-iscsi/8dba0dd : FAILURE in 28m 01s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/9/check/fujitsu-eternus-dx-fc/0496cc0 : FAILURE in 29m 09s","accounts_in_message":[],"_revision_number":9},{"id":"c2dc1edf1664668f4d6372ef79c8525d0686e9e9","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-04-23 03:46:55.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9008a676dac71b100459402e9aaaa4e13c6dff44","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-23 04:03:03.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/70/851170/9/check/iscsi-driver-alletra-6k/fe888d3/fe888d3.html : FAILURE in 46m 04s","accounts_in_message":[],"_revision_number":9},{"id":"cd756d26df9dfc0d640fabb911a6d0c5ba11bdde","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-04-23 04:03:30.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/851170 : FAILURE in 45m 21s","accounts_in_message":[],"_revision_number":9},{"id":"a2af6593c9c7466fde9b24d3679ff71cb539d460","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-04-23 04:10:31.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/902d06fffd5740b7ac30fbd473f52cd5\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/b5eccfb451c149359c8ef2b67549d5d3 : FAILURE in 47m 51s","accounts_in_message":[],"_revision_number":9},{"id":"cfced96a4f4a0394c9b6d0bbfc9b85e22968e68f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-04-23 04:13:07.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/70/851170/9/ : FAILURE in 49m 08s","accounts_in_message":[],"_revision_number":9},{"id":"4b330bcb01da9216b6ebc18e12988a1773d930e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-23 05:09:40.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/09ac344e5a6e4d648d87a05ec906c214\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7243cc4b89f24556b3f1db07835232d9 : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca958b09dc4949b49c2baefd44ac81cc : SUCCESS in 13m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fa1e6c1f7d44865af3215a5cabc0689 : SUCCESS in 12m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4204365ee1b14cdfa6fb68e804c3d491 : SUCCESS in 15m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/364dce092b1b4683b9db081bd1f26342 : SUCCESS in 44m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1b31a8ac6f9a48f9a35c5c84e06f591c : SUCCESS in 1h 32m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/549541e61140465cb95d3f4fe20b9e07 : SUCCESS in 1h 14m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83c0778f8a184da9ace8afe87d9fe5b7 : SUCCESS in 9m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bc06d1d496764865ba4ac698eb2e2b82 : SUCCESS in 14m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d4826fca0c6a45a4b8a40509e7c08eb7 : SUCCESS in 5m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c0b2adcb7b8c402e9aa43508381a9149 : FAILURE in 6m 35s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d76e40706b4e46b9a4b45568cea1ff21 : SUCCESS in 6m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c811a5f0b624459baec1b6261a4c5b1f : SUCCESS in 5m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/98b278c0b0ad401dadfc8766532814d3 : SUCCESS in 53m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e800cb18c89b40e6b22c6467a553c4d8 : SUCCESS in 6m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/036ae6d358b7407ab663f5c3b352bc1f : POST_FAILURE in 10m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c5509b9aa8364c82a1ff83af71443f39 : POST_FAILURE in 13m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/64c12ca3aa3a4bb58096641b263a627c : SUCCESS in 1h 51m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/15bbb45f43ac4da1830c636fa89c3de0 : SUCCESS in 1h 46m 45s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0efa080c56a84f2cba51612c648912da : SUCCESS in 50m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4c918b94be034feaad0c3ec40640255a : SUCCESS in 1h 33m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f07533714c224098bb1a011fe1e2bd16 : SUCCESS in 1h 00m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e6c404fadd4b41d29e5dadfaac66bfbc : SUCCESS in 1h 35m 16s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0b22a585ae784beab88cb05a4ac3f173 : SUCCESS in 1h 30m 21s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/55c1914e46034cd6ad15a3eae805acab : SUCCESS in 1h 39m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71eaae8d29e346b4bf996177354099fd : SUCCESS in 43m 48s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7cd00c3a44045fa91709515cdcfc68e : FAILURE in 12m 04s","accounts_in_message":[],"_revision_number":9},{"id":"ab1ad91aa59a9f5dddf3c45d826001f9ceae86f1","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-04-23 05:11:40.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"07e3049050f4919fae0e690977363006786598ec","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-04-23 05:41:44.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/9/check/fujitsu-eternus-dx-iscsi/770f6ae : FAILURE in 28m 38s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/9/check/fujitsu-eternus-dx-fc/0410389 : FAILURE in 29m 29s","accounts_in_message":[],"_revision_number":9},{"id":"c665e577ea11924cfc248b602cd57255e6537494","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-04-23 07:41:59.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/70/851170/9/check/tempest-dsvm-full-focal-py3/7bac05f/ : FAILURE in 46m 50s\n- devstack-plugin-ceph-dsvm-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/70/851170/9/check/devstack-plugin-ceph-dsvm-focal-py3/29fe5cd/ : FAILURE in 51m 00s","accounts_in_message":[],"_revision_number":9},{"id":"6272338083f5d7258beb08d453b4c163ed0ba70e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-04-23 12:40:15.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/9/20230423190022/NEC-V-ISCSI : SUCCESS in 2h 31m 52s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/70/851170/9/20230423190022/NEC-V-FC : SUCCESS in 2h 39m 49s","accounts_in_message":[],"_revision_number":9},{"id":"234b4e28850a8a4d68ed8b3e46f7af6cbd5d5b84","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-04-24 10:50:23.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/70/851170/9/check/DellEMC_Unity_FC/738026f/DellEMC_Unity_FC/62 : FAILURE in 30m 54s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/70/851170/9/check/DellEMC_Unity_iSCSI/6e62465/DellEMC_Unity_iSCSI/62 : FAILURE in 19m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"25a963bf699ba3d81a5cd640e7f5201a6903c21e","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-04-25 07:23:13.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6cc580df78eeb79232a37ddc900b7c902995802a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-04-25 21:34:46.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-70-851170-9/ns5_nfs : FAILURE in 20m 1s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-70-851170-9/ns5_iscsi : FAILURE in 19m 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"909bfe9c285b9abea69a819683ef0caa40c95a7c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-04-25 23:00:04.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-851170-9/FC/ : SUCCESS in 2h 37m 23s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-70-851170-9/iSCSI/ : SUCCESS in 2h 28m 25s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"f9b3de43e7f575fae00aa3530d724f7284850886","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-05-16 02:46:37.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"03bc297fc5628467c929aee0b1cc261b29982490","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 02:58:47.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f998237485c4d75bbc4c682cf8cf411\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b23ec42808744f4ca8cf59f360fb8a0f : SUCCESS in 10m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/45997feb7a52448783150dd3755e105d : SUCCESS in 10m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/119bba8416714f439803436f233759b5 : SUCCESS in 10m 09s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"dddbcb659864c91e4fc661a32f58278ba64edc2e","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-05-16 03:21:40.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d6c2d4afca5fbf1bd775108769e1c65b235d14d3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-16 03:24:35.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2fddeba5b462402eb62b5f706d57dbea : RETRY_LIMIT in 11m 42s","accounts_in_message":[],"_revision_number":9},{"id":"c486c4f43eb16ce80f039236693898e815110bfe","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-16 03:36:57.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/0da88e86cae543b6aed262274eb2e845 : POST_FAILURE in 49m 37s","accounts_in_message":[],"_revision_number":9},{"id":"b37b6aabc56e2b2114984d9d46fcc761b279efbe","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-16 03:40:39.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/1646e3f17a9f4dfea2881d73c28b8efc\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/264e184e7f0c433b90dc0776444b8044 : FAILURE in 50m 59s","accounts_in_message":[],"_revision_number":9},{"id":"cb5177bfbffe8bff1f1735ad5799513727d6b627","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 04:50:58.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2ea572064f641d48091f3c4da7f68bc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f622620a6b748c78d746fc04380e7f3 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ff00861a70d4c64ad19c75e8b040005 : SUCCESS in 12m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f82ca01856f94b64a953020e2fba70d5 : SUCCESS in 15m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c482a7384d64b608868d4842add1740 : SUCCESS in 14m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57be55e2f2684bae96a4360ed88146b1 : SUCCESS in 16m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/9cbe07dd8ac54d8fa9bd8bf026a2fdd5 : SUCCESS in 1h 07m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/593c30d365b84f919216d031fcbf5fad : SUCCESS in 1h 43m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/743d62fa863e45319c7c1b9a767d72ef : SUCCESS in 1h 09m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9abc8f5c13442df8275fcfcc19b6a8e : SUCCESS in 9m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b0ab17d4689e4e0f8820514f27acbc92 : SUCCESS in 17m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2242e8b1d4df4bf09cd8133b0f2d4fe5 : SUCCESS in 5m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8c5b370990eb41cf938780d4a4a11b7e : FAILURE in 6m 15s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f36d04c425ff4a67a464c70cf35f866c : SUCCESS in 6m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/27e92ebb783c4cf388a3ac1ef7e0ade7 : SUCCESS in 5m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ef4a753dc74643f090ddc214e83e940b : SUCCESS in 48m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5511cdb1bdae465b8326afafaebef065 : SUCCESS in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e4cc9fee10144c5998c5b1a402f13fe3 : FAILURE in 1h 23m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/25bd0f3116f34661a569b262292ebdd1 : FAILURE in 1h 46m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f5f979984a6b441b89925e8c368ff85a : SUCCESS in 1h 52m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e415e876a12e4261b44078a538a04ac9 : SUCCESS in 1h 54m 33s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/69eaa335e901451284e191b14ecaf381 : SUCCESS in 1h 13m 41s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/22d8b38049eb47db96f4a11bb05f94ff : SUCCESS in 1h 28m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/378830bdd8084d77b20a7a8bd99a038d : SUCCESS in 48m 23s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5a2cf817610b4c9eb5fdb16ff012c545 : SUCCESS in 1h 01m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2deb65a6ff7f40b8bb19b0aa7847ed5e : SUCCESS in 1h 40m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/370d282430ca428aae8fae71b601981c : SUCCESS in 2h 00m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6b61db198f845d2a0905579fbc45e72 : SUCCESS in 59m 28s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71316d3f469e4a888339136af05b9711 : SUCCESS in 1h 23m 05s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/d371a111ba2f49bd8b570d2ad5b9f531 : SUCCESS in 12m 45s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8ca3e3ca614a493c10898d65059d237a37c4ae8e","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-05-16 05:01:35.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/9/check/fujitsu-eternus-dx-iscsi/c4bde7d : FAILURE in 25m 17s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/9/check/fujitsu-eternus-dx-fc/c5b9bb1 : SUCCESS in 1h 39m 17s","accounts_in_message":[],"_revision_number":9},{"id":"fc359e64fa7a2e0d2480f360475cfe14feeb424b","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-05-18 02:03:37.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"40997a8738e642cc4ec76aa1e950659024417f2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 02:26:24.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad31bf7d69ba4db7b0753a778002f8ec\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/52f151121a9641f79b950fcd129dd6eb : SUCCESS in 13m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c2b05f39e46a460badc90b9390719e45 : SUCCESS in 14m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/49ad7c04fff8403e8d5f938b43ed4a3b : SUCCESS in 12m 30s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8f1686ea100d13cbb5ad922fea1c68c65c06b388","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-18 02:31:52.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/439619d0c52e448bb363dfca5ae8d146\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/45ac2966df4c4435aafcb3c66f8ef49a : FAILURE in 26m 20s","accounts_in_message":[],"_revision_number":9},{"id":"ab0c7ee71d85465ecd584239355a9c5351d6516c","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-18 02:50:24.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/68b9963a047d44ab9c27717c58aab9d6 : POST_FAILURE in 46m 05s","accounts_in_message":[],"_revision_number":9},{"id":"914262d1ba83f558219cef97731bfd10999435cc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-18 03:05:47.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/21814391e3f74b9598d950d70c4e1b95 : SUCCESS in 1h 00m 28s","accounts_in_message":[],"_revision_number":9},{"id":"676d561114fed6de4154977c678d2b37cfe8b588","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-05-18 03:55:58.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/9/check/fujitsu-eternus-dx-iscsi/08483e4 : SUCCESS in 1h 51m 41s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/9/check/fujitsu-eternus-dx-fc/8f04a61 : FAILURE in 34m 53s","accounts_in_message":[],"_revision_number":9},{"id":"e7e3023ec6c115e5ed99165479c618897fb25cba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 05:19:16.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73fc8546da5b418da0afb166896adb3b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dff7ff218f084954a78a731c138389b5 : SUCCESS in 8m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3d36e5eccb946ada3119d1c336d9155 : SUCCESS in 12m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b505ed3c662e40d68debf94586d6af2d : SUCCESS in 12m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d532433dea54eddbdfe2765f80f3207 : SUCCESS in 13m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fee96a5f1d7b4abd99c806b8139cff8c : SUCCESS in 13m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/0d7d807998e0487aa95c7eea5db18d92 : SUCCESS in 1h 03m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/943c96ab049142c69e0b382e0cea346d : SUCCESS in 1h 27m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a3669a07620442989fd56616b605b35 : SUCCESS in 1h 16m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd7ae428af004c0a8d97e7f4b9335acc : SUCCESS in 9m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fd31c4d20dfc48b1ad89271fe87b870e : SUCCESS in 19m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1f9e8a3de4304598ab81345f7d8d0c20 : SUCCESS in 5m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ce83dc3bdbd843b5a1273e7e9d6de5c7 : FAILURE in 6m 17s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cfb659b35b2f4858b90306e1e05eb6e9 : SUCCESS in 5m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c55fb1b764ba4963b57d6d073ca2d63b : SUCCESS in 5m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4838c9254a8f479f951f8fa870963eeb : SUCCESS in 47m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/431b683a7d2746038b2da39392589f8d : SUCCESS in 6m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6bd091fa315b487eaf8e9351f31e7057 : SUCCESS in 1h 38m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fda6d8df33a84af1ad1a54edfa6f234e : FAILURE in 1h 51m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eaeeb774697c46c49eafcae156bd8df5 : SUCCESS in 2h 02m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/64debe5960c9439b92db53b68501d610 : POST_FAILURE in 20m 41s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1f87f4bee79f4f4ebb4ba8e36528ac83 : SUCCESS in 1h 20m 46s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4ba371cc8d9e46d2b49b7b5ec4f4a1a2 : SUCCESS in 1h 23m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5a0c84c1a8d74ffca01d874f3bc31200 : SUCCESS in 54m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2eb2e7ce87b24cfe876290990d5fb41a : SUCCESS in 1h 40m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/82dd1386d8df4e70a0e05315f451fa07 : FAILURE in 36m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4d5a2cc091b946bca8d07fda10ef913f : SUCCESS in 2h 44m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a46debdba5e4494583db48e5ea777623 : SUCCESS in 1h 14m 54s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d096e839f8e54adba5e42e6f7d37237c : SUCCESS in 1h 16m 02s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/2ec13c9aeeaa414abbc8cfdff9596eac : SUCCESS in 11m 26s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"931caf3e6e7db3baf1e1e05cea2c3a3deaa7845c","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-05-18 05:21:11.000000000","message":"Patch Set 9:\n\nBuild failed.","accounts_in_message":[],"_revision_number":9},{"id":"7193af829f9ebe08dffa172a3083c3463946d68a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-05-18 05:25:13.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/70/851170/9/check/DellEMC_PowerFlex/1e6dca3/DellEMC_PowerFlex/91 : FAILURE in 5m 35s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"a4621e667da654687dc9daead968307fe273ce12","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-05-18 05:26:30.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_05_2023_05_19_27\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":9},{"id":"980efad896a44ac10d6d67a9769e5c97f65ef5ed","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-05-18 05:44:44.000000000","message":"Patch Set 9:\n\nBuild failed.","accounts_in_message":[],"_revision_number":9},{"id":"6613b6c1880ab381063cd5e8526aa926730ea02b","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-05-18 05:45:07.000000000","message":"Patch Set 9:\n\nBuild failed.","accounts_in_message":[],"_revision_number":9},{"id":"8b94118b528bf6f3d089cce6a83d5fa7e164eca7","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-05-18 05:55:57.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c0941394abdc1173e4089e7b83e8f181ae04c465","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-18 06:31:04.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/851170/9/upstream-check/cinder-tempest-plugin-ontap-fcp/e961d63/ : RETRY_LIMIT in 10m 27s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/851170/9/upstream-check/cinder-tempest-plugin-ontap-iscsi/ec3d793/ : SUCCESS in 45m 10s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/851170/9/upstream-check/cinder-tempest-plugin-ontap-nfs/b45c8c2/ : SUCCESS in 46m 46s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/70/851170/9/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/43baf15/ : SUCCESS in 1h 09m 25s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"a0999a6622cc17e058ab57159cec5cb4eb5ad8da","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-18 07:19:53.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/70/851170/9/check/DellEMC_Unity_FC/8b44911/DellEMC_Unity_FC/153 : SUCCESS in 1h 34m 21s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/70/851170/9/check/DellEMC_Unity_iSCSI/c52b8f7/DellEMC_Unity_iSCSI/156 : SUCCESS in 2h 00m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"ddb7b4083df2987fe5a634d5e44ca8b11361ec51","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-05-18 07:23:46.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/9/check/fujitsu-eternus-dx-iscsi/b7e3c33 : SUCCESS in 1h 27m 10s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/9/check/fujitsu-eternus-dx-fc/d149538 : FAILURE in 12m 55s","accounts_in_message":[],"_revision_number":9},{"id":"2c8b66088e9d47dc93b5d110def2bff0f494bdb9","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-05-18 08:28:10.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"509a0832e0440a85cc324be06d2f86e2ea5bc9b3","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-05-18 10:03:29.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/9/check/fujitsu-eternus-dx-iscsi/fab7261 : FAILURE in 1h 34m 28s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/9/check/fujitsu-eternus-dx-fc/f42a188 : FAILURE in 26m 39s","accounts_in_message":[],"_revision_number":9},{"id":"9f317ee2c9af95eb6dccc57039ccb9bab3e38abe","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-05-18 11:25:48.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/70/851170/9/check/DellEMC_PowerMAX_FC/c5da8ff/DellEMC_PowerMAX_FC/68 : FAILURE in 6h 06m 10s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/70/851170/9/check/DellEMC_PowerMAX_iSCSI/c63ed47/DellEMC_PowerMAX_iSCSI/67 : FAILURE in 5h 43m 53s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":9},{"id":"c76733c9d01b028b915cd1830d1b9a30e9b44143","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-05-19 14:50:01.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/36f7bc1b33fb4bcf86964b33ae08cc72 : RETRY_LIMIT in 1m 04s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/e788f65c0b2041078b9edcbe1116daaa : RETRY_LIMIT in 1m 11s","accounts_in_message":[],"_revision_number":9},{"id":"5ce97dee9f25387f11e42b98d2e320c1a7dd0d0f","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-05-22 08:40:39.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"882e20bac35dba413c7a07e5c1a92b444938664e","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-05-22 08:43:57.000000000","message":"Patch Set 9:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/9/check/fujitsu-eternus-dx-iscsi/6433837 : FAILURE in 1m 07s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/9/check/fujitsu-eternus-dx-fc/7b70d82 : ABORTED in 2m 36s","accounts_in_message":[],"_revision_number":9},{"id":"7f8c7ef509a8fd67ed50112c47b9e4b0a266e8fa","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-05-22 08:46:15.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3bff1be6598b5c2b0d5a171639eb652400098b0c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-05-22 10:32:21.000000000","message":"Patch Set 9:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/70/851170/9/check/fujitsu-eternus-dx-iscsi/70c54ff : SUCCESS in 1h 34m 37s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/70/851170/9/check/fujitsu-eternus-dx-fc/680af10 : SUCCESS in 1h 45m 28s","accounts_in_message":[],"_revision_number":9},{"id":"0bb1e05d33c1ae5c4be34ac12af7d2f606d2d4cf","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-05-22 18:10:52.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/70/851170/9/check/DellEMC_PowerStore_iSCSI/88a7291/DellEMC_PowerStore_iSCSI/47 : FAILURE in 6m 09s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/70/851170/9/check/DellEMC_PowerStore_FC/22213a0/DellEMC_PowerStore_FC/48 : SUCCESS in 1h 06m 29s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/70/851170/9/check/DellEMC_PowerStore_NFS/5a5b3e5/DellEMC_PowerStore_NFS/46 : FAILURE in 6m 05s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/70/851170/9/check/DellEMC_PowerStore_NVMeTCP/1a9cf85/DellEMC_PowerStore_NVMeTCP/46 : FAILURE in 5m 57s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"f14dffaa3798607fedf340b9f9f19f6035315d0f","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-09-14 16:49:08.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"aa2f558eb15e2f5fb857813d52710edcf461a2f7","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-09-14 21:00:06.000000000","message":"Patch Set 9:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/70/851170/9/fc : SUCCESS in 0h 48m 56s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/70/851170/9/iscsi : SUCCESS in 0h 46m 12s","accounts_in_message":[],"_revision_number":9},{"id":"2a672d042ea3cab1b7c54b35b0a1717398b48867","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-09-15 01:14:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"7ea7e7e0d7acd13d8b839decf2153e2ee99852e6","revisions":{"a07e41bcbf4984c3313313a484619a98a9c5974a":{"kind":"REWORK","_number":1,"created":"2022-07-27 07:41:59.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/70/851170/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/851170/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/851170/1"}}},"commit":{"parents":[{"commit":"25043468d3921a23eadd31df47f3d98918465ebf","subject":"Merge \"Fix mypy job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25043468d3921a23eadd31df47f3d98918465ebf"}]}],"author":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-27 07:45:40.000000000","tz":-240},"committer":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-27 07:45:40.000000000","tz":-240},"subject":"Fujitsu Driver: Add manage related support","message":"Fujitsu Driver: Add manage related support\n\nAdded supported for manage related function in Fujitsu driver\n\nAdded support for the following functions:\\n* get manageable volume\\n* get manageable snapshot\\n* manage volume\\n* manage snapshot\n\nChange-Id: I03575e97f9e57685480b62d8b14060262428c8ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a07e41bcbf4984c3313313a484619a98a9c5974a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a07e41bcbf4984c3313313a484619a98a9c5974a"}]},"branch":"refs/heads/master"},"2c2a41bc0e6741e629bca3d7125a6bf05436048e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-27 07:44:16.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/70/851170/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/851170/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/851170/2"}}},"commit":{"parents":[{"commit":"25043468d3921a23eadd31df47f3d98918465ebf","subject":"Merge \"Fix mypy job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25043468d3921a23eadd31df47f3d98918465ebf"}]}],"author":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-27 07:45:40.000000000","tz":-240},"committer":{"name":"XuQi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-27 07:44:16.000000000","tz":0},"subject":"Fujitsu Driver: Add manage related support","message":"Fujitsu Driver: Add manage related support\n\nAdded supported for manage related function in Fujitsu driver\n\nAdded support for the following functions:\n* get manageable volume\n* get manageable snapshot\n* manage volume\n* manage snapshot\n\nChange-Id: I03575e97f9e57685480b62d8b14060262428c8ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c2a41bc0e6741e629bca3d7125a6bf05436048e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c2a41bc0e6741e629bca3d7125a6bf05436048e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5f64cf9fd7bb9e5634e5d27ee1f9125837fd7d20":{"kind":"REWORK","_number":3,"created":"2022-12-06 02:51:46.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/70/851170/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/851170/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/851170/3"}}},"commit":{"parents":[{"commit":"25043468d3921a23eadd31df47f3d98918465ebf","subject":"Merge \"Fix mypy job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25043468d3921a23eadd31df47f3d98918465ebf"}]}],"author":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-27 07:45:40.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2022-12-06 02:55:42.000000000","tz":-300},"subject":"Fujitsu Driver: Add manage related support","message":"Fujitsu Driver: Add manage related support\n\nAdded supported for manage related function in Fujitsu driver\n\nAdded support for the following functions:\n* get manageable volume\n* get manageable snapshot\n* manage volume\n* manage snapshot\n\nChange-Id: I03575e97f9e57685480b62d8b14060262428c8ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f64cf9fd7bb9e5634e5d27ee1f9125837fd7d20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5f64cf9fd7bb9e5634e5d27ee1f9125837fd7d20"}]},"branch":"refs/heads/master"},"ea77b64eb1a809551fd3cba1daeddc987415425c":{"kind":"REWORK","_number":4,"created":"2022-12-06 09:37:48.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/70/851170/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/851170/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/851170/4"}}},"commit":{"parents":[{"commit":"25043468d3921a23eadd31df47f3d98918465ebf","subject":"Merge \"Fix mypy job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25043468d3921a23eadd31df47f3d98918465ebf"}]}],"author":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-27 07:45:40.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2022-12-06 09:41:52.000000000","tz":-300},"subject":"Fujitsu Driver: Add manage related support","message":"Fujitsu Driver: Add manage related support\n\nAdded supported for manage related function in Fujitsu driver\n\nAdded support for the following functions:\n* get manageable volume\n* manage volume\n\nChange-Id: I03575e97f9e57685480b62d8b14060262428c8ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ea77b64eb1a809551fd3cba1daeddc987415425c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ea77b64eb1a809551fd3cba1daeddc987415425c"}]},"branch":"refs/heads/master"},"8bfef1ef64ce1c9b6daae5112203a67755f2a8d1":{"kind":"REWORK","_number":5,"created":"2022-12-07 04:00:43.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/70/851170/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/851170/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/851170/5"}}},"commit":{"parents":[{"commit":"25043468d3921a23eadd31df47f3d98918465ebf","subject":"Merge \"Fix mypy job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25043468d3921a23eadd31df47f3d98918465ebf"}]}],"author":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-27 07:45:40.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2022-12-07 04:05:22.000000000","tz":-300},"subject":"Fujitsu Driver: Add manage related support","message":"Fujitsu Driver: Add manage related support\n\nAdded supported for manage related function in Fujitsu driver\n\nAdded support for the following functions:\n* get manageable volume\n* manage volume\n\nChange-Id: I03575e97f9e57685480b62d8b14060262428c8ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8bfef1ef64ce1c9b6daae5112203a67755f2a8d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8bfef1ef64ce1c9b6daae5112203a67755f2a8d1"}]},"branch":"refs/heads/master"},"a5dc416809dfede07a5214fe74b4464faec535c1":{"kind":"REWORK","_number":6,"created":"2022-12-10 08:21:16.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/70/851170/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/851170/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/851170/6"}}},"commit":{"parents":[{"commit":"25043468d3921a23eadd31df47f3d98918465ebf","subject":"Merge \"Fix mypy job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25043468d3921a23eadd31df47f3d98918465ebf"}]}],"author":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-27 07:45:40.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2022-12-10 08:25:41.000000000","tz":-300},"subject":"Fujitsu Driver: Add manage related support","message":"Fujitsu Driver: Add manage related support\n\nAdded supported for manage related function in Fujitsu driver\n\nAdded support for the following functions:\n* get manageable volume\n* manage volume\n\nChange-Id: I03575e97f9e57685480b62d8b14060262428c8ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a5dc416809dfede07a5214fe74b4464faec535c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a5dc416809dfede07a5214fe74b4464faec535c1"}]},"branch":"refs/heads/master"},"f437245b7f7a42a9dca125a017ca8680a88309c4":{"kind":"REWORK","_number":7,"created":"2022-12-12 08:47:49.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/70/851170/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/851170/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/851170/7"}}},"commit":{"parents":[{"commit":"25043468d3921a23eadd31df47f3d98918465ebf","subject":"Merge \"Fix mypy job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25043468d3921a23eadd31df47f3d98918465ebf"}]}],"author":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-27 07:45:40.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2022-12-12 08:52:23.000000000","tz":-300},"subject":"Fujitsu Driver: Add manage related support","message":"Fujitsu Driver: Add manage related support\n\nAdded supported for manage related function in Fujitsu driver\n\nAdded support for the following functions:\n* get manageable volume\n* manage volume\n\nChange-Id: I03575e97f9e57685480b62d8b14060262428c8ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f437245b7f7a42a9dca125a017ca8680a88309c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f437245b7f7a42a9dca125a017ca8680a88309c4"}]},"branch":"refs/heads/master"},"98b304377254f6f4184a9c0d446f423f9a09ecd9":{"kind":"REWORK","_number":8,"created":"2023-04-14 05:22:48.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/70/851170/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/851170/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/851170/8"}}},"commit":{"parents":[{"commit":"78fac2960d28e97f42454bb5f385b46fbae23a3c","subject":"Merge \"Shut up pylint about win32_disk_size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/78fac2960d28e97f42454bb5f385b46fbae23a3c"}]}],"author":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-27 07:45:40.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2023-04-14 05:26:34.000000000","tz":-240},"subject":"Fujitsu Driver: Add manage volume support","message":"Fujitsu Driver: Add manage volume support\n\nAdded supported for manage volume function in Fujitsu driver\n\nAdded support for the following functions:\n* get manageable volume\n* manage volume\n\nChange-Id: I03575e97f9e57685480b62d8b14060262428c8ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/98b304377254f6f4184a9c0d446f423f9a09ecd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/98b304377254f6f4184a9c0d446f423f9a09ecd9"}]},"branch":"refs/heads/master"},"7ea7e7e0d7acd13d8b839decf2153e2ee99852e6":{"kind":"REWORK","_number":9,"created":"2023-04-23 03:15:35.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/70/851170/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/70/851170/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/70/851170/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/70/851170/9"}}},"commit":{"parents":[{"commit":"78fac2960d28e97f42454bb5f385b46fbae23a3c","subject":"Merge \"Shut up pylint about win32_disk_size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/78fac2960d28e97f42454bb5f385b46fbae23a3c"}]}],"author":{"name":"Xu Qi","email":"xuq.fnstxz@fujitsu.com","date":"2022-07-27 07:45:40.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2023-04-23 03:20:30.000000000","tz":-240},"subject":"Fujitsu Driver: Add manage volume support","message":"Fujitsu Driver: Add manage volume support\n\nAdded supported for manage volume function in Fujitsu driver\n\nAdded support for the following functions:\n* get manageable volume\n* manage volume\n\nChange-Id: I03575e97f9e57685480b62d8b14060262428c8ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ea7e7e0d7acd13d8b839decf2153e2ee99852e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ea7e7e0d7acd13d8b839decf2153e2ee99852e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
