)]}'
{"id":"openstack%2Fcinder~852454","triplet_id":"openstack%2Fcinder~stable%2Fyoga~I9368c914169fd457d2826968cdde077f38055a30","project":"openstack/cinder","branch":"stable/yoga","topic":"flake8-logging","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-08-10 19:33:38.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9368c914169fd457d2826968cdde077f38055a30","subject":"DNM: checking flake8-logging-format\u003d\u003d0.7.4 damage","status":"ABANDONED","created":"2022-08-08 16:19:02.000000000","updated":"2022-08-10 19:33:38.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7e3b6f5d6ea6e3adccb15170a4c0a9ba5b2bf7ee","_number":852454,"virtual_id_number":852454,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-08-08 16:39:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-08 16:19:10.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-08-08 16:19:22.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-08 16:19:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-08 16:19:23.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-08 16:19:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-08 16:19:37.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-08-08 16:19:51.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-08 16:20:24.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-08-08 18:44:42.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-08-08 18:45:50.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-08-08 21:13:30.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-08-09 14:56:26.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"}],"messages":[{"id":"aa82d853fd6d6f2888ba016d2f642bc82132dcec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-08 16:19:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85fe9314db6dcd26bef13ef81d489a9dadc3eec8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-08 16:19:10.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1659975549129 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"a063e65282145308b6bcef32a439b4eedff3074d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-08 16:19:22.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch stable/yoga.  The error was:\n\n  found undefined alias \u0027gate-irrelevant-files\u0027\n  in \"openstack/cinder/.zuul.yaml@stable/yoga\", line 7, column 31","accounts_in_message":[],"_revision_number":1},{"id":"362e0ddb39769cfddbe7713567d52d03542b6d45","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-08 16:19:22.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch stable/yoga.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@stable/yoga\", line 7, column 31","accounts_in_message":[],"_revision_number":1},{"id":"2de480a0684605f97ade7ee2f1bbce1d241cff4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 16:19:23.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch stable/yoga.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@stable/yoga\", line 7, column 31","accounts_in_message":[],"_revision_number":1},{"id":"e074816c0f6d343fbd65c661b1bbebe8ab6c1cd2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-08-08 16:19:23.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch stable/yoga.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@stable/yoga\", line 7, column 31","accounts_in_message":[],"_revision_number":1},{"id":"4cd659c3f1c867b2c483641dea8e38d6474cdbbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 16:19:31.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch stable/yoga.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@stable/yoga\", line 7, column 31","accounts_in_message":[],"_revision_number":1},{"id":"e9e8fa8a102a08a960f4d82e584d51da58ceba7f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-08 16:19:37.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch stable/yoga.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@stable/yoga\", line 7, column 31","accounts_in_message":[],"_revision_number":1},{"id":"9c3f61fcec7497a5d8123f14a21c58506d1f7e78","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-08 16:19:44.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch stable/yoga.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@stable/yoga\", line 7, column 31","accounts_in_message":[],"_revision_number":1},{"id":"630a5938c55fcf6a3eb75b76e46f1bd275197563","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-08 16:19:51.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch stable/yoga.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@stable/yoga\", line 7, column 31","accounts_in_message":[],"_revision_number":1},{"id":"f5de3d3da321a6d3050a6e804684e59376da3d7f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-08 16:20:24.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Unable to update gerrit/openstack/cinder","accounts_in_message":[],"_revision_number":1},{"id":"dda7d36d499aec4e018f185e0291be50249c32a3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-08 16:21:10.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/cinder on branch stable/yoga.  The error was:\n\n  found undefined alias\n  in \"openstack/cinder/.zuul.yaml@stable/yoga\", line 7, column 31","accounts_in_message":[],"_revision_number":1},{"id":"0971ef5d03738844bd43ee6c801dbe75d01b3468","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-08 16:23:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"511c55f169fa25649984167f936d33e70ac2fc65","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-08 16:23:35.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1659975814133 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"e6e5785a508810b48b2f608ae0b8c95435aa739b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-08 16:24:44.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Unable to update gerrit/openstack/cinder","accounts_in_message":[],"_revision_number":2},{"id":"d20922e98b8f792f23354b3af178da59858e39d7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-08 16:37:41.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/54/852454/2/check/dsvm-tempest-cisco-zonemanager-job/1922414/ : FAILURE in 9m 10s","accounts_in_message":[],"_revision_number":2},{"id":"5093a7cdb2c2bf61e36701ad4592823c445c6535","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 16:39:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba5aba22203a4aedabe0a0b4ee593374\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/869241b53fe148dda2bbf24772fb2e19 : SUCCESS in 9m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ea6b17c2ef04ddc95ee59e8e4709d28 : SUCCESS in 14m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07d698fb622e414aa9abadb47f7c3bd5 : SUCCESS in 14m 24s","accounts_in_message":[],"_revision_number":2},{"id":"4c59953ea9fa90162669faf497d16e0d61363c38","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-08 17:14:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bb05d726912f07af4b0a3406ef84c6ac0652b4f6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-08 17:16:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/54/852454/2/check/iscsi-driver-alletra-6k/c64c014/c64c014.html : FAILURE in 52m 25s","accounts_in_message":[],"_revision_number":2},{"id":"26b61e19e8171cc27c2227389bba8032ebacdbd9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-08 17:24:47.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/08725f1163e447f3ac6ed6675276b17c : POST_FAILURE in 55m 44s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/4b241510c5404c1cb3a47d012ab75a36 : POST_FAILURE in 55m 56s","accounts_in_message":[],"_revision_number":2},{"id":"a637e8ddc06536f686c0002e984919bfd9866dd0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-08 17:44:02.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm-focal https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/852454 : FAILURE in 1h 02m 39s","accounts_in_message":[],"_revision_number":2},{"id":"7134ad15714de8d9606698b604a76cf23bcaf7c9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-08 18:44:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/54/852454/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/aa59bfd/ : SUCCESS in 1h 44m 09s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/54/852454/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d7caac6/ : SUCCESS in 1h 26m 04s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/54/852454/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/45ece90/ : FAILURE in 32m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"414b66888d071b5bde52ee5f1884a1785ea2b628","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-08 18:45:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/852454/2/iscsi/ : SUCCESS in 1h 34m 49s\n- cinder-smb http://cloudbase-ci.com/cinder/852454/2/smb/ : SUCCESS in 1h 07m 22s","accounts_in_message":[],"_revision_number":2},{"id":"a2abbee4cfb87d9556e2b1a6bc3e00b2d8409ef2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-08 21:13:30.000000000","message":"Patch Set 2:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/08_08_2022_19_16_35\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"9813daa688af62e2b91323f17a7dd1a6cd14273b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-08 21:50:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/852454/2/check/iscsi-driver-nimble/b033e6f/b033e6f.html : FAILURE in 49m 56s","accounts_in_message":[],"_revision_number":2},{"id":"e7d5f9115849ce832cc363745740e6c41a4707de","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-08-09 14:56:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/852454/2/20220809211018/NEC-V-FC : SUCCESS in 2h 10m 38s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/852454/2/20220809211018/NEC-V-ISCSI : SUCCESS in 2h 46m 05s","accounts_in_message":[],"_revision_number":2},{"id":"7e3b6f5d6ea6e3adccb15170a4c0a9ba5b2bf7ee","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-10 19:33:38.000000000","message":"Abandoned\n\nNo longer needed.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"248c8150c28bb75366015b6879c896f964887957","revisions":{"77fe8a949ed20ca7baef7d094250bbafc19267dc":{"kind":"REWORK","_number":1,"created":"2022-08-08 16:19:02.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/54/852454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/852454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/852454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/852454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/852454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/852454/1"}}},"commit":{"parents":[{"commit":"9b583401adfd63d3d311bb9515f9a27b9b679a6f","subject":"SF: Remove compat clone image code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9b583401adfd63d3d311bb9515f9a27b9b679a6f"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-08-08 16:17:46.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-08-08 16:17:46.000000000","tz":-240},"subject":"DNM: checking flake8-logging-format\u003d\u003d0.7.4 damage","message":"DNM: checking flake8-logging-format\u003d\u003d0.7.4 damage\n\nChange-Id: I9368c914169fd457d2826968cdde077f38055a30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77fe8a949ed20ca7baef7d094250bbafc19267dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77fe8a949ed20ca7baef7d094250bbafc19267dc"}]},"branch":"refs/heads/stable/yoga"},"248c8150c28bb75366015b6879c896f964887957":{"kind":"REWORK","_number":2,"created":"2022-08-08 16:23:23.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/54/852454/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/852454/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/852454/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/852454/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/852454/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/852454/2"}}},"commit":{"parents":[{"commit":"9b583401adfd63d3d311bb9515f9a27b9b679a6f","subject":"SF: Remove compat clone image code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9b583401adfd63d3d311bb9515f9a27b9b679a6f"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-08-08 16:17:46.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-08-08 16:23:11.000000000","tz":-240},"subject":"DNM: checking flake8-logging-format\u003d\u003d0.7.4 damage","message":"DNM: checking flake8-logging-format\u003d\u003d0.7.4 damage\n\nChange-Id: I9368c914169fd457d2826968cdde077f38055a30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/248c8150c28bb75366015b6879c896f964887957"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/248c8150c28bb75366015b6879c896f964887957"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
