)]}'
{"id":"openstack%2Fcinder~853100","triplet_id":"openstack%2Fcinder~stable%2Fyoga~Idef086c55885687684875603481582f763355885","project":"openstack/cinder","branch":"stable/yoga","topic":"bug/1934435-stable/yoga","attention_set":{},"removed_from_attention_set":{"30555":{"account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"last_update":"2022-08-22 19:46:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idef086c55885687684875603481582f763355885","subject":"NetApp SolidFire: Fix request errors while Element OS upgrade","status":"MERGED","created":"2022-08-15 17:07:17.000000000","updated":"2022-08-22 20:31:41.000000000","submitted":"2022-08-22 19:46:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853100-bug/1934435-stable/yoga","cherry_pick_of_change":802957,"cherry_pick_of_patch_set":2,"meta_rev_id":"04919a4d07da1d1d7bdc25caa132400db424e909","_number":853100,"virtual_id_number":853100,"owner":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-22 19:46:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":2,"date":"2022-08-15 17:13:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-19 15:53:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-19 15:53:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-15 17:07:17.000000000","updated_by":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"reviewer":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"state":"CC"},{"updated":"2022-08-15 17:07:28.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-08-15 17:13:10.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2022-08-15 17:21:50.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-08-15 17:25:36.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-08-15 17:33:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-15 17:43:05.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-15 18:13:03.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-08-15 18:48:46.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-08-15 19:46:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-15 20:03:16.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-08-15 23:29:28.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-08-18 20:45:09.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-19 15:53:34.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2022-08-22 20:31:41.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"eba5373a99f8925598a615fdfccd82e4fe822000","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2022-08-15 17:07:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a9f1720a2b85cf672f1d4dcce724e0be70bf6b13","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-15 17:07:28.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1660583245357 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"2c481218abaf3b39c18ac5b4d5ca315db7f137c3","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-08-15 17:13:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1d31ae5c86cb00e184a0f0c4115c19b2b01af85a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-15 17:21:50.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/853100/1/iscsi/ : FAILURE in 13m 53s\n- cinder-smb http://cloudbase-ci.com/cinder/853100/1/smb/ : FAILURE in 13m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e4a63407d3cd991d11b8fe8e2d52030fee8cf76f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-15 17:25:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/00/853100/1/check/dsvm-tempest-cisco-zonemanager-job/4b3ebb0/ : POST_FAILURE in 11m 22s","accounts_in_message":[],"_revision_number":1},{"id":"eae1b9ab64570d50f6d345681d04c173581e948b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 17:33:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea440f2696af4e87aab5a9ac920f7393\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bb99a391e58645329ea4efb0fd92e3a9 : SUCCESS in 12m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/56bf30e15fc94ed689310dd067839210 : SUCCESS in 24m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d6178c7a0f132ba13a1b74201e74347730385686","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-15 17:43:05.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/853100/1/check/tempest-dsvm-full-focal-py3/4286672/ : FAILURE in 33m 06s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/853100/1/check/devstack-plugin-ceph-dsvm-focal-py3/1f94e97/ : FAILURE in 32m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a0653df6a3c50922eb960b1b64906e9e440dff14","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-15 18:13:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/00/853100/1/check/iscsi-driver-alletra-6k/1fffd32/1fffd32.html : FAILURE in 59m 21s","accounts_in_message":[],"_revision_number":1},{"id":"e3f2854a0f371cb81e4034527e3d650ca7a78c4b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-15 18:48:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/00/853100/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/39dc932/ : SUCCESS in 1h 31m 12s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/00/853100/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/91e37f9/ : SUCCESS in 1h 40m 10s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/00/853100/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/29ee939/ : FAILURE in 36m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b006827486a94c5d65ded6aa8362a3bbad7ecf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 19:46:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73adef116f6f414ba1c9300302092db6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba79e410676c40cba4e2ef09f6a9d80d : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51ca2f735b5545a5aa4a2af9ace6d1e8 : SUCCESS in 11m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c5cb3f4da7248e794f365d0d7ba4f40 : SUCCESS in 12m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74e662eb52ea4457bfb9fae650d072a0 : SUCCESS in 12m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/f28771e8aedc475a9a4355a6c43d1052 : SUCCESS in 1h 02m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ca1c329adc7b4601a5280a3cc1e95eb1 : SUCCESS in 1h 06m 58s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/91c86a43868f4dc8972dc282028e5439 : SUCCESS in 1h 21m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe3f21bb1bc64ab3a9b027231b85a835 : SUCCESS in 51m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbcb8c3a17f0459cafd36f3ff8345c43 : SUCCESS in 8m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/47d2bf892afe4d6591585883fea081f4 : SUCCESS in 14m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3adc5bb971024800a31d2ecefcdeac2f : SUCCESS in 6m 24s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fa3ea53a642841cc9c1970413a5cd175 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/564e0b11ff2b4708a46888b6cc6ba050 : SUCCESS in 7m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/048c21e00cf24eb68ce23d2206238b19 : SUCCESS in 7m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/19fce34488614069a3edf22f7597f63e : SUCCESS in 42m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/426e4ee67d914ee394baefc52699c5d0 : FAILURE in 6m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2ae9e75af68e46de9f3d4c276d726c33 : SUCCESS in 1h 29m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8670ef9582c048d7b81bd997d3991dc8 : FAILURE in 1h 32m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/71d15c61ea00400eaba306ad39f1bde2 : SUCCESS in 1h 56m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/655d1c5bb71549d59f2b9d3c2b0112be : SUCCESS in 1h 11m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d1a7ff36dcc14ea88cb07d62fd7e34ac : SUCCESS in 1h 26m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/de61002985e04d4eadd2ef52fa221b41 : SUCCESS in 44m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/63f49e24022848068779ce168de413df : SUCCESS in 1h 24m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/912e319614814ed7a6e255d4f2e60f70 : SUCCESS in 2h 34m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4000c8f25d9849a6b3ff890784582819 : SUCCESS in 55m 40s","accounts_in_message":[],"_revision_number":1},{"id":"00569769c858f5dc334ba97ecd0ae999d4ee7a14","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-08-15 20:03:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-00-853100-1/1660592802 : SUCCESS in 16m 32s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":1},{"id":"1307c71075cabcb6429db64096307382bcef76fc","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-08-15 23:29:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/853100/1/20220816024027/NEC-V-FC : SUCCESS in 5h 05m 34s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/853100/1/20220816024027/NEC-V-ISCSI : SUCCESS in 5h 48m 58s","accounts_in_message":[],"_revision_number":1},{"id":"51574329b6358ede8bfcf1fc7b99f014e6b73391","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-08-18 20:45:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/853100/1/20220819020911/NEC-FC : SUCCESS in 3h 32m 43s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/853100/1/20220819020911/NEC-ISCSI : SUCCESS in 3h 35m 54s","accounts_in_message":[],"_revision_number":1},{"id":"dc14e3f172997f99261562285b614fe41768f048","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-08-19 15:53:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aad12de035222ba26bc507db202dcd19250837ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 15:54:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fed52a03e3d725f40a1a5c0420d71fa3056470e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 17:22:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed4dcfb09d134b04896844f48d5c573a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6387f20587874677869ad071e2f3fc56 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d6872fabb204bd085dec8eae868dd52 : SUCCESS in 10m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bce36aadf03948b19b07d9e7afb085da : SUCCESS in 12m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25db8978a709458bb275e03c35043949 : SUCCESS in 11m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/fbebc29f70274acd8a05081e75d9031a : SUCCESS in 57m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/44a21cd12e0d4a5e91bbe90d52f0dfb9 : SUCCESS in 1h 22m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00dd2089d7a347c0af893ebed44f397d : SUCCESS in 49m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9097bc9b3fa48cdadb1e8cef12781ff : SUCCESS in 8m 35s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e6850293b9e14822826262ad972f101d : SUCCESS in 1h 03m 41s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9274b1c488e944a8bfc7f03a288d7bf8 : FAILURE in 1h 20m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69e86a2587d54de6a517c2a988553b3b : SUCCESS in 57m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f45248835a3208baf2353099f7549bbf8343df0b","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-08-22 15:24:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3f61e0fed5fd9e8c5edc1e5ad48c8a5acbd3272","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 15:51:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b72345d835cd43f9923d1d2bc7b5e6c2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/33f34ed053914eff96ae25c98c0c0759 : SUCCESS in 20m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/97f7b7f149f746d9a046b99570d03db4 : FAILURE in 20m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"27b43d0e1a4c49cfbf8fe38599fce184704925e1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-22 16:07:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/00/853100/1/check/dsvm-tempest-cisco-zonemanager-job/deb9f0b/ : POST_FAILURE in 12m 00s","accounts_in_message":[],"_revision_number":1},{"id":"40c8dce2b8a204de1d659dc5c398001d97f5aefb","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-22 16:43:42.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/853100/1/iscsi/ : SUCCESS in 1h 03m 06s\n- cinder-smb http://cloudbase-ci.com/cinder/853100/1/smb/ : SUCCESS in 1h 18m 49s","accounts_in_message":[],"_revision_number":1},{"id":"b9b726187cbe5887f8c504f04a8140d699b967ce","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-22 17:08:37.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/853100/1/check/tempest-dsvm-full-focal-py3/bfa4255/ : FAILURE in 1h 07m 53s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/00/853100/1/check/devstack-plugin-ceph-dsvm-focal-py3/65e82b9/ : FAILURE in 1h 13m 54s","accounts_in_message":[],"_revision_number":1},{"id":"027929aaae5c9a0935d4aa58e5b932509c7f888f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 18:07:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7581f63b0d174e4fbfa77cde0450ad07\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/393e1348d34d47af8fc80f29758feec5 : SUCCESS in 8m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/395fcb010a0e48c08dcfa054a9f5f7b2 : SUCCESS in 13m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b889a8d92359477996f12c4d94bbafcc : SUCCESS in 12m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96266340bdd74e0c9d5059830930b232 : SUCCESS in 13m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/8e886d59726e47139daab391d44cd29d : SUCCESS in 55m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7a4bd9d0f6c54208b6e4f8a8ae231ddd : SUCCESS in 1h 06m 49s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ca45f2484a424eae8d539a5936501560 : SUCCESS in 1h 29m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/325ad0152c3c486d83152632d6e63842 : SUCCESS in 48m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1fc1d0e077d409987d9e0449303cc40 : SUCCESS in 7m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/723103b790e04f43999428116dcbe606 : SUCCESS in 14m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d637d2a0be7c48298c5a9aa4ec9ba387 : SUCCESS in 6m 13s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/de6f73f7c65a4c2b8e67c074262e8482 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f2a66168e3154572932334bef2277da6 : SUCCESS in 7m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fba61cff69804bf29a7aa774ed55a0aa : SUCCESS in 8m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/923e28ef74c64f46922ba7d9e75449fe : SUCCESS in 50m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/70b1e5efaecc4927bad82db0ddafaa1b : FAILURE in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/933d294963974e948e7226123950719e : SUCCESS in 1h 33m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/76de6bd5e18849bcaccf274029790bf3 : FAILURE in 1h 48m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/89c533198d6f4c5ea88e5abc3efb9683 : SUCCESS in 1h 58m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f62b29809a754954909285b0fea593f9 : SUCCESS in 1h 15m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/61f43d181437490bbf463e161ee12ed9 : FAILURE in 2h 11m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/53acc661fd9048fbbf746736f09c2813 : SUCCESS in 43m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3820d6e71b1e46679e9e79cc69871637 : SUCCESS in 1h 26m 23s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0f1cdd5329e64674945b3a5f988e11fe : SUCCESS in 2h 34m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41a1d7b9655340ee901f8a05e8d9f884 : SUCCESS in 1h 08m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c9174d021c0ee765690b5049da1883f09ca06929","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 18:08:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7318d36d0329d8f4ba54d65d645bbf60a42f73f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 19:46:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/407e1a9843da47648a90893def506b82\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36e848a02a6d4c9794aacad1cba01520 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d9762b6915e4e6c8fab232a960e08ff : SUCCESS in 13m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20e22f2659164433a9851f20cb4c8ee9 : SUCCESS in 15m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b33f9f53e2fb4bd89bc8c49389d9bb4e : SUCCESS in 14m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/51eb9d52d2804f34a7aea2e85e61d8d9 : SUCCESS in 1h 07m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0b8a9ebe2b064abead5e7d9b1e68f16e : SUCCESS in 1h 36m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7cdad8e7cda43618a4181113a7e9f92 : SUCCESS in 59m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe10866956424b7787ae38cd33a12a5c : SUCCESS in 11m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/94222e6a9e69434d9d10f7f779e31d4e : SUCCESS in 1h 17m 53s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3269a62150a74fd9a0d10a9b58c26019 : SUCCESS in 1h 25m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8583c2e999b340548fc3275a78e584c8 : SUCCESS in 1h 01m 59s","accounts_in_message":[],"_revision_number":1},{"id":"021d255e8682786edfc681bb804fe0c166ca39d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 19:46:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"87283e256896da90b5c5fe48cd07ddfffedc3fb0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 19:47:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7f9df35eff84b60afe9f959be1fd1c3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/246fbf0156f64ab28b725330db25611d : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb12ee56c7db4ffc9b037fc0155dd186 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1},{"id":"04919a4d07da1d1d7bdc25caa132400db424e909","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-22 20:31:41.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/22_08_2022_20_12_45\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"44947ae1e96c865641cc9d58bb17cb2cb0182eaa","revisions":{"44947ae1e96c865641cc9d58bb17cb2cb0182eaa":{"kind":"REWORK","_number":1,"created":"2022-08-15 17:07:17.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/00/853100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/853100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/853100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/853100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/853100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/853100/1"}}},"commit":{"parents":[{"commit":"9b583401adfd63d3d311bb9515f9a27b9b679a6f","subject":"SF: Remove compat clone image code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9b583401adfd63d3d311bb9515f9a27b9b679a6f"}]}],"author":{"name":"Fabio Oliveira","email":"fabioaurelio1269@gmail.com","date":"2021-07-21 18:31:51.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2022-08-15 17:07:17.000000000","tz":0},"subject":"NetApp SolidFire: Fix request errors while Element OS upgrade","message":"NetApp SolidFire: Fix request errors while Element OS upgrade\n\nWhen Element OS is being upgraded, the requests received will\nget exceptions as result, but the operation will normally succeed\na few attempts after.\n\nThis patch fixes this issue by adding the following exceptions to\nour list of retryable exceptions in SolidFire driver:\n    - xDBOperationTimeout.\n    - xDBConnectionLoss.\n    - xNoHandler.\n    - xSnapshotFailed.\n    - xRecvTimeout.\n    - xDBNoSuchPath.\n    - xPermissionDenied.\n\nChange-Id: Idef086c55885687684875603481582f763355885\nCloses-Bug: #1934435\n(cherry picked from commit b7b830e9ac1e58ef432edf892eb972c98817ffdf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44947ae1e96c865641cc9d58bb17cb2cb0182eaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/44947ae1e96c865641cc9d58bb17cb2cb0182eaa"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
