)]}'
{"id":"openstack%2Fcinder~853325","triplet_id":"openstack%2Fcinder~master~I2139cf0be019d852ecf716b20d63181a65204b63","project":"openstack/cinder","branch":"master","topic":"unit-test-delays","attention_set":{},"removed_from_attention_set":{"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2022-08-17 15:14:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2139cf0be019d852ecf716b20d63181a65204b63","subject":"Inspur: : Eliminate timed delays in unit tests","status":"MERGED","created":"2022-08-16 16:28:15.000000000","updated":"2022-08-29 01:40:35.000000000","submitted":"2022-08-17 15:14:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853325-unit-test-delays","meta_rev_id":"0cd2af0e325bf8c40ba928a9c7818ce7e37dae0d","_number":853325,"virtual_id_number":853325,"owner":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-17 15:14:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-08-16 17:40:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2022-08-17 13:36:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-08-17 13:36:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-16 16:28:17.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-08-16 16:28:27.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-08-16 16:43:44.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-08-16 16:43:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-16 16:48:26.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-08-16 16:52:49.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-08-16 17:01:36.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-08-16 17:24:31.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-08-16 17:40:43.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-08-16 18:35:26.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-08-16 18:44:25.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-16 18:49:04.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-08-16 21:08:40.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2022-08-16 21:13:57.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-16 22:05:25.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-08-16 22:39:25.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-08-16 22:59:22.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2022-08-17 13:36:20.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2022-08-17 15:17:02.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-08-17 16:21:54.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-08-29 01:40:35.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"35e66caec37947732d1bddae8d5de5efd286422e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-08-16 16:28:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98dc542e8ad64ac8aa82abfd490bda85a83245b7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-16 16:28:17.000000000","message":"Patch Set 1:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":1},{"id":"67bb37bb532ecd7890b1aa5a1eb6fe2d186d9abb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-16 16:28:27.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1660667302625 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"eb0babc8c2a69261b109bb3d3f5e8e638b650a48","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-16 16:43:44.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/853325/1/iscsi/ : FAILURE in 13m 48s\n- cinder-smb http://cloudbase-ci.com/cinder/853325/1/smb/ : FAILURE in 13m 51s","accounts_in_message":[],"_revision_number":1},{"id":"dd594abd7a117fc1bad63507d1342729056a5070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 16:43:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38f06a9ed2a04ebca648c396f7e8de9d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddbc6976a20245a8ae032fb517352ace : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b85327fe31b54727b2c96bcb54953b9f : SUCCESS in 11m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ddc873c0b5c4ce680dcdce14be468eb : SUCCESS in 10m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf67d23a00b546b0a0965813391c8d97 : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/542d8fa9d8714cad82a22fc3cd80e224 : SUCCESS in 10m 59s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/05e054c1316c445eb8a6ccd8d6a6bb09 : SUCCESS in 13m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b2fabac5272648bc8ee5a940f6684ccc : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":1},{"id":"7d53a1a883ad26460cbb91bcb8dbc547b0d444da","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-16 16:48:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k finger://openstackcifc-virtual-machine/2c976cc0023745bd9544d2dcc5693b2e : RETRY_LIMIT in 10s","accounts_in_message":[],"_revision_number":1},{"id":"b99d27f112dbb243b0f702ae3b87921142019baa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 16:52:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5d30951b50b4164bf6df7599a510780\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9d8ade45bd94c2ba9d8a22a302222ae : SUCCESS in 15m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f75eac0eb7a44a599be9161f82e6ff85 : SUCCESS in 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8239ab134ebe2a076c959d29cb1a56c8264453bc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-16 16:52:49.000000000","message":"Patch Set 1:\n\nBuild SUCCESS. log path: https://github.com/LightBitsLabs/logs/tree/main/16_08_2022_16_44_11\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"2887be62f79072d093289b7f25d7c6bbd51e1d6a","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-08-16 17:01:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-25-853325-1/1660668239 : SUCCESS in 17m 35s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":1},{"id":"31eefc260218a91313aa0641d13e54418aaf19a5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-16 17:24:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/25/853325/1/check/dsvm-tempest-cisco-zonemanager-job/6b6b61c/ : POST_FAILURE in 11m 37s","accounts_in_message":[],"_revision_number":1},{"id":"f2fed509e75322722a520c54e0d191752355aa55","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-08-16 17:35:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-25-853325-1/1660668739 : SUCCESS in 16m 49s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":1},{"id":"c7cebd8df83017a21fdfc97ee70759a219c51901","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-16 17:40:43.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb08f3d2644fc5ed5c4ceafdffc254e97688b728","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-16 18:35:26.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/25/853325/1/check/DellEMC_PowerStore_FC/c965d86/DellEMC_PowerStore_FC/742 : SUCCESS in 1h 20m 02s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/25/853325/1/check/DellEMC_PowerStore_NFS/9a0496f/DellEMC_PowerStore_NFS/283 : SUCCESS in 1h 30m 47s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/25/853325/1/check/DellEMC_PowerStore_NVMeTCP/47fa73e/DellEMC_PowerStore_NVMeTCP/157 : SUCCESS in 1h 50m 55s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"95f92615d050e7f1033d66c90b293c39af5828c0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-16 18:44:25.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/25/853325/1/check/DellEMC_PowerFlex/4fae35d/DellEMC_PowerFlex/1172 : ABORTED in 2h 00m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"fc444874157d3d58b22fa99099616a088e266347","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-08-16 18:49:04.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/25/853325/1/check/DellEMC_SC_FC/9c57749/DellEMC_SC_FC/929 : SUCCESS in 52m 22s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/25/853325/1/check/DellEMC_SC_iSCSI/193e3bd/DellEMC_SC_iSCSI/947 : SUCCESS in 2h 04m 42s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"8314257bd83030cb9003d285ec996eee4da89b32","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-08-16 21:08:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/853325-1/HPMSA-iSCSI/ : SUCCESS in 55:53.99\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/853325-1/HPMSA-FC/ : SUCCESS in 52:02.92","accounts_in_message":[],"_revision_number":1},{"id":"c249c9dde48180f3e69ff7a431b19f0c486ba15e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-16 21:13:57.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-25-853325-1/ns5_nfs : SUCCESS in 3h 27m 26s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-25-853325-1/ns5_iscsi : SUCCESS in 3h 33m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"310a02f992da15003a3553be7b664d45fa10bc35","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-08-16 22:05:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/853325-1/Seagate-iSCSI/ : FAILURE in 1:08:44\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/853325-1/Seagate-FC/ : SUCCESS in 1:09:55","accounts_in_message":[],"_revision_number":1},{"id":"c8824f0f25ef6d0fa50f38ad236e384aa2ba6e35","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-08-16 22:39:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/853325-1/PVME-iSCSI/ : SUCCESS in 1:07:27\n- PVME-FC https://os-logs.seagate.com/dell-ci/853325-1/PVME-FC/ : SUCCESS in 1:05:01","accounts_in_message":[],"_revision_number":1},{"id":"c50b6d2e76d20d99bf84d85af24cee5ffe65f892","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-08-16 22:59:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/853325-1/Lenovo-iSCSI/ : SUCCESS in 1:07:36\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/853325-1/Lenovo-FC/ : SUCCESS in 1:05:53","accounts_in_message":[],"_revision_number":1},{"id":"e505df372184c5704cf96a46741f042b2b4fd733","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-08-17 13:36:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ce0d74f4f7c057ac7f117988536df2253b7a9083","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 13:37:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58989668d27096a4ae246d32edc5ac5ac90f8602","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 15:14:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab1b502ec85647e49b4efb88e038dc8b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/318ea53ee01f4a8e906ba8f3965d84d3 : SUCCESS in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cba78cabdce24bb5bde72901b488eca9 : SUCCESS in 12m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffb83775e77944c094fadf94d745b79e : SUCCESS in 11m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2970d9ed055c4d6689418d87f2be4231 : SUCCESS in 12m 47s","accounts_in_message":[],"_revision_number":1},{"id":"76d9052db3a1b4d0853c1adddbab1591b10291e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 15:14:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cf56933ba8d407467f515db18b39d49cc89fc7b6","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-08-17 15:17:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/25/853325/1/20220817214703/NEC-V-FC : SUCCESS in 2h 00m 21s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/25/853325/1/20220817214703/NEC-V-ISCSI : SUCCESS in 2h 29m 55s","accounts_in_message":[],"_revision_number":1},{"id":"671c3ac6448fdedde361778a999c7ef49fe652a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 15:17:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3571e0c33b604c4ba7cc4a9a947e5897\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1989e6f7c52b4306a2eb64bfb72ba893 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1},{"id":"60888e43e28ecb6ec91425afb776e2b6794180ac","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-17 16:21:54.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/25/853325/1/check/DellEMC_Unity_FC/26bd6ec/DellEMC_Unity_FC/1470 : FAILURE in 21m 36s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/25/853325/1/check/DellEMC_Unity_iSCSI/b3bbc30/DellEMC_Unity_iSCSI/None : RETRY_LIMIT\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"0cd2af0e325bf8c40ba928a9c7818ce7e37dae0d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-08-29 01:40:35.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aef79bcbd481da0e86992328d080f83aa450ba2b","revisions":{"aef79bcbd481da0e86992328d080f83aa450ba2b":{"kind":"REWORK","_number":1,"created":"2022-08-16 16:28:15.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/25/853325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/25/853325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/853325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/853325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/25/853325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/25/853325/1"}}},"commit":{"parents":[{"commit":"f3c1ed28d20065e63eda9497081cc495a743d628","subject":"Veritas: Eliminate timed delays in unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f3c1ed28d20065e63eda9497081cc495a743d628"}]}],"author":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2022-08-16 14:46:06.000000000","tz":-420},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2022-08-16 16:24:27.000000000","tz":-420},"subject":"Inspur: : Eliminate timed delays in unit tests","message":"Inspur: : Eliminate timed delays in unit tests\n\nEliminate unit test delays by using \"zero interval\" looping calls.\n\nChange-Id: I2139cf0be019d852ecf716b20d63181a65204b63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aef79bcbd481da0e86992328d080f83aa450ba2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/aef79bcbd481da0e86992328d080f83aa450ba2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
