)]}'
{"id":"openstack%2Fcinder~854266","triplet_id":"openstack%2Fcinder~master~I44511900c7e8434398bf9467e5067fc1dfc0940d","project":"openstack/cinder","branch":"master","topic":"rbd-native-treads","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-08-30 01:47:18.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-09-13 07:45:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I44511900c7e8434398bf9467e5067fc1dfc0940d","subject":"Tests: Fix zone manager tests","status":"MERGED","created":"2022-08-23 20:10:44.000000000","updated":"2022-09-15 14:49:11.000000000","submitted":"2022-09-13 07:45:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"854266-rbd-native-treads","meta_rev_id":"eff0eb972225079a1a67d8cd0d06f058bd547bf1","_number":854266,"virtual_id_number":854266,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-13 07:45:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-08-30 01:47:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2022-08-29 19:50:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2022-08-30 11:24:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-13 07:29:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-13 07:29:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"}],"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-23 20:10:57.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-08-23 20:32:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-24 01:07:31.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-08-24 01:25:31.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-08-24 02:09:59.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-24 02:25:28.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-24 03:38:36.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-24 05:46:39.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-08-24 08:40:26.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-08-24 08:55:17.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-08-24 14:41:45.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-24 21:00:23.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-08-24 21:31:16.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2022-08-24 21:51:30.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-08-25 00:35:35.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-08-25 02:07:04.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2022-08-25 02:46:30.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-08-25 08:24:50.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-08-26 04:49:15.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-29 01:39:17.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-29 19:50:38.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2022-08-30 11:24:04.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-09-01 19:13:49.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-09 23:59:15.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-09-13 07:29:17.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-09-15 14:49:11.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"ee6c85289e1d4323b1dff2804d13e66834b98b3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-23 20:10:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6b971057ba51605191ccc5e2d6be463cd348fa4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-23 20:10:57.000000000","message":"Patch Set 1:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":1},{"id":"9d355d91997bb3e0fdd6a43ce030c3ea5988d0ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 20:32:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/737823477fc047528ea32355b1c6dc61\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3b0aae03673491aa9007cb366965bbd : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41b0651de34d4dd0ac1a94e7deafb83e : SUCCESS in 10m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3afba63416a04c88bea88e569ce1cf6a : SUCCESS in 13m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16525b66d6934b4dab96c306979722e9 : SUCCESS in 16m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee382f7171784c6fa6a0f9100268dcb0 : SUCCESS in 12m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2b901024025d491ea6b5ff4bc4ca9ccd : SUCCESS in 19m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f5c1a8310ec94e21b139f51ecbbd3d7a : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"5c27911b01e806e71a49d110f0318b7e5add731a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 21:14:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/467acfc8eb1c46ebadd892421d44bfa5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c417351a5ec841059e20850bb1485722 : SUCCESS in 37m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f2ca5d31c21b4ae0bf09031542f5c1d2 : SUCCESS in 22m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"edba379cdbeaa4bf5809ef5c19b179e5bb7bdb41","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-24 01:07:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/66/854266/1/check/iscsi-driver-alletra-6k/16251d4/16251d4.html : FAILURE in 1h 04m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8cb5cf6ee106955d3d8331703185516b1ff32cd4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-24 01:25:31.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661285456421 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"2ca96a51cfa6ded1d203b27f00339f58fc491ed1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-24 02:09:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/854266 : SUCCESS in 1h 23m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9bad756a7b77bde07ac28183871bb396f5bc7f00","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-24 02:25:28.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-66-854266-1/ns5_nfs : SUCCESS in 3h 33m 14s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-66-854266-1/ns5_iscsi : SUCCESS in 3h 36m 5s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"01198007a6ab713e0ba91f13ba48dcf2009bbd18","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-24 03:38:36.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/66/854266/1/check/fujitsu-eternus-dx-iscsi/36c8e4b : SUCCESS in 1h 41m 21s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/66/854266/1/check/fujitsu-eternus-dx-fc/3275930 : SUCCESS in 1h 50m 16s","accounts_in_message":[],"_revision_number":1},{"id":"79e6218724e648b0cce77da497724ca401033da9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-24 05:46:39.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/854266/1/iscsi/ : FAILURE in 1h 37m 19s\n- cinder-smb http://cloudbase-ci.com/cinder/854266/1/smb/ : SUCCESS in 1h 03m 52s","accounts_in_message":[],"_revision_number":1},{"id":"f2bc1ec35040bd2f64462e0f77c7769983f66573","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-24 08:40:26.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/24_08_2022_08_19_03\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"3465dc35c90741f5476120e1b8be9cefb21d2189","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-24 08:55:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/66/854266/1/check/dsvm-tempest-cisco-zonemanager-job/26a46ab/ : POST_FAILURE in 12m 58s","accounts_in_message":[],"_revision_number":1},{"id":"dbc062d2e9a60de3c86aae4ac60f4dfb204f3367","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-24 14:41:45.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/66/854266/1/check/DellEMC_PowerFlex/5a1ee0e/DellEMC_PowerFlex/1203 : ABORTED in 2h 00m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"7d0519f22681c23df9d7862dc2a6cb9a557bd6aa","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-24 21:00:23.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/66/854266/1/check/DellEMC_PowerStore_FC/5475432/DellEMC_PowerStore_FC/771 : SUCCESS in 1h 20m 53s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/66/854266/1/check/DellEMC_PowerStore_NFS/69651dd/DellEMC_PowerStore_NFS/311 : SUCCESS in 1h 33m 37s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/66/854266/1/check/DellEMC_PowerStore_NVMeTCP/13077e7/DellEMC_PowerStore_NVMeTCP/186 : FAILURE in 1h 47m 32s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"e1a2fceda453392e84dc6f544ffbdac9684d4aff","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-08-24 21:31:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/854266-1/HPMSA-iSCSI/ : SUCCESS in 1:06:21\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/854266-1/HPMSA-FC/ : SUCCESS in 53:57.41","accounts_in_message":[],"_revision_number":1},{"id":"3d7b8a83179cf914ba08d96d1f817c4cef10a9a9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-24 21:51:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f3376882f87dd3ed5c83ad230f016ff575e24c9e","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-08-25 00:35:35.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/854266-1/PVME-iSCSI/ : SUCCESS in 1:07:01\n- PVME-FC https://os-logs.seagate.com/dell-ci/854266-1/PVME-FC/ : SUCCESS in 1:06:50","accounts_in_message":[],"_revision_number":1},{"id":"93a14f11d8f428d161a4f595812684c8234659be","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-08-25 02:07:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/854266-1/Lenovo-iSCSI/ : SUCCESS in 1:07:00\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/854266-1/Lenovo-FC/ : SUCCESS in 1:02:08","accounts_in_message":[],"_revision_number":1},{"id":"e8026a13676c99f987584a0566a4b0fbe8e047cb","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-08-25 02:46:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/854266-1/Seagate-iSCSI/ : SUCCESS in 1:13:14\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/854266-1/Seagate-FC/ : SUCCESS in 1:09:47","accounts_in_message":[],"_revision_number":1},{"id":"2f81e3402a02a61a21cfb6811a660633c8a03183","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-25 08:24:50.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/66/854266/1/check/DellEMC_Unity_FC/6c8bf00/DellEMC_Unity_FC/1531 : FAILURE in 14m 23s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/66/854266/1/check/DellEMC_Unity_iSCSI/e0561ef/DellEMC_Unity_iSCSI/1419 : SUCCESS in 1h 51m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"717a55c62c94bd1d4f83196f81e4c6ee2321b3f1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-08-26 04:49:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/66/854266/1/20220826101743/NEC-FC : SUCCESS in 3h 29m 06s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/66/854266/1/20220826101743/NEC-ISCSI : SUCCESS in 3h 31m 28s","accounts_in_message":[],"_revision_number":1},{"id":"8603f44a1188cd6a9732e1fc8a33583303ec5282","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-08-29 01:39:17.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"7d77c35adc2baa4d8ea995e7a087b8d345ca6c15","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-29 12:22:09.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c09f5a3c49a5c9f6d0bbcf73fe2e08a293c6b20c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-29 15:06:56.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"4c6e33dabad6d331542bb696401c2c8311a090f0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-29 15:07:06.000000000","message":"Patch Set 2:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2},{"id":"8a639c5408c31166d91f5d92a4bd9cc2a6eb7c56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-29 15:10:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ab61ae70c39deeffe10fbc078a7f57556ef42aa5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-29 15:10:23.000000000","message":"Patch Set 3:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":3},{"id":"eb832930cd7e7edb9a1d49a58bc5221db71b247e","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-29 15:10:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ebaa6346b06b20450136915550e97a2b8a70b371","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 15:32:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb401219586a447b96008fc7b9f27641\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a6feb2614b84dc0bc5c9d2cf6daecd6 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d146e7ee4de4613a4cab24b4f88981f : SUCCESS in 13m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51e7dd0a74d84db5bb341613dabd120d : SUCCESS in 13m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b875876b56d34a0f808069b467ab1d59 : SUCCESS in 10m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbcf6214520e4fff9c518e45a360e864 : SUCCESS in 15m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b7175b729f5f450a8f59b74e4975e238 : SUCCESS in 16m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/94122c46cb35462fa89304883633bbb0 : SUCCESS in 7m 58s","accounts_in_message":[],"_revision_number":3},{"id":"badd6b47bb6e61a206c99773f76a020a4cb69f5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 16:30:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d0ac0f185c64fb4a74d702f3a1b18a4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/98ccadcc0db24f1f8eb1dacae5d73363 : SUCCESS in 21m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05219320e20642c48b1af0490684e4ec : SUCCESS in 21m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"791b9e412120b618aac14caf99e15297bae1f274","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-29 17:31:50.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661785628424 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"2f3d79a41dce673cf2732e32e92e7364e6b1263b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-29 18:48:21.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/854266 : FAILURE in 36m 54s","accounts_in_message":[],"_revision_number":3},{"id":"46342abd835de12b556a5bb659a87c0923706f21","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-29 19:41:32.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/854266 : SUCCESS in 1h 25m 48s","accounts_in_message":[],"_revision_number":3},{"id":"2518f1582c30a4e4e539567ef9b8ce35090d3af6","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2022-08-29 19:50:38.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f9603eb1cbc692b5a0fa9bf871d87a7567a58f47","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-29 20:13:05.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/66/854266/3/check/iscsi-driver-alletra-6k/a9e90f5/a9e90f5.html : FAILURE in 1h 36m 09s","accounts_in_message":[],"_revision_number":3},{"id":"9d130281869f208a4b7ec4fb1c89e464db7ea121","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-29 20:34:35.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-66-854266-2/ns5_nfs : SUCCESS in 3h 34m 7s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-66-854266-2/ns5_iscsi : SUCCESS in 3h 32m 0s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"26f8dbf9f979dc3a27bbf8188b5c50cb48792b3a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-29 22:03:10.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/854266/3/iscsi/ : SUCCESS in 59m 14s\n- cinder-smb http://cloudbase-ci.com/cinder/854266/3/smb/ : SUCCESS in 1h 19m 59s","accounts_in_message":[],"_revision_number":3},{"id":"befd7a898562d00159082b1c0e9375d184a5134a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-30 01:16:02.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/66/854266/3/check/DellEMC_PowerFlex/7ebc613/DellEMC_PowerFlex/1238 : ABORTED in 2h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"5312e845532fd8c0da8213f19cb32903376fd285","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-30 01:47:18.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4563734bbdf125423fd1f047d8d7fd04166f3ff4","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-30 01:49:54.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/66/854266/3/check/DellEMC_PowerStore_FC/21fa356/DellEMC_PowerStore_FC/808 : SUCCESS in 1h 19m 06s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/66/854266/3/check/DellEMC_PowerStore_NFS/e5ec287/DellEMC_PowerStore_NFS/346 : SUCCESS in 1h 29m 38s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/66/854266/3/check/DellEMC_PowerStore_NVMeTCP/dfc7e52/DellEMC_PowerStore_NVMeTCP/221 : SUCCESS in 1h 51m 23s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"bf410f2a8fc05597a51c3a1bb61844fbd6d114a4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-30 02:41:20.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-66-854266-3/ns5_nfs : SUCCESS in 3h 31m 34s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-66-854266-3/ns5_iscsi : SUCCESS in 3h 40m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"ad96fc4af561823d6288a9c10b8154b3d1c83fbc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-30 03:47:35.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661785828437 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"ae9b2792f65ee87d5b43e21844920fa50f4eece9","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-08-30 04:04:10.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/854266-3/HPMSA-iSCSI/ : SUCCESS in 56:01.42\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/854266-3/HPMSA-FC/ : SUCCESS in 53:59.34","accounts_in_message":[],"_revision_number":3},{"id":"3c4b56017c9aa290300724a7e807afae603dabd2","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-08-30 06:06:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/854266-3/PVME-iSCSI/ : SUCCESS in 1:08:44\n- PVME-FC https://os-logs.seagate.com/dell-ci/854266-3/PVME-FC/ : SUCCESS in 1:07:19","accounts_in_message":[],"_revision_number":3},{"id":"88239954285e8966efc160770db63af7b046a777","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-08-30 07:10:35.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/854266-3/Lenovo-iSCSI/ : SUCCESS in 1:15:04\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/854266-3/Lenovo-FC/ : SUCCESS in 1:13:09","accounts_in_message":[],"_revision_number":3},{"id":"3921267021fbb553f41c68cf3e4ae999d45a32cb","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-08-30 08:39:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/854266-3/Seagate-iSCSI/ : FAILURE in 1:13:51\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/854266-3/Seagate-FC/ : SUCCESS in 1:11:23","accounts_in_message":[],"_revision_number":3},{"id":"092bedf867012a8442ca20c304d643867068abf8","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-08-30 11:24:04.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"106e1e48d4c21261bba032951a34b4fd7f64f7f9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-30 13:33:53.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/66/854266/3/check/DellEMC_Unity_FC/2e61fea/DellEMC_Unity_FC/1602 : FAILURE in 14m 43s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/66/854266/3/check/DellEMC_Unity_iSCSI/244c1ae/DellEMC_Unity_iSCSI/1469 : FAILURE in 2h 21m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"8e752fa063e9a2f187e595dc30d0fa9903c65574","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-30 14:04:08.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/66/854266/3/check/dsvm-tempest-cisco-zonemanager-job/fbbbb01/ : SUCCESS in 1h 05m 00s","accounts_in_message":[],"_revision_number":3},{"id":"09bab684eb7cd9ac89f6beddfd9d7cbf83da9cb7","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-09-01 19:13:49.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/66/854266/3/ : SUCCESS in 1h 28m 26s","accounts_in_message":[],"_revision_number":3},{"id":"a2e6977ce98fc070dbb903e653adf3e204791302","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-09-02 21:27:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/66/854266/3/20220903025519/NEC-ISCSI : SUCCESS in 3h 27m 07s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/66/854266/3/20220903025519/NEC-FC : SUCCESS in 3h 31m 43s","accounts_in_message":[],"_revision_number":3},{"id":"9fd91a251a187e4c1bb2f09e217cf12205d2557e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-09-09 23:59:15.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/66/854266/3/20220910034338/NEC-V-ISCSI : SUCCESS in 2h 19m 04s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/66/854266/3/20220910034338/NEC-V-FC : SUCCESS in 5h 15m 33s","accounts_in_message":[],"_revision_number":3},{"id":"aba8fb0c03f050b389d3a0acc06cb959bece6381","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-09-13 07:29:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8f7647404b99c1b624cc8e39813448eb07bee457","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 07:29:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"43760497a35faed912fd776827b310395b8ceb0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 07:45:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5da37c4293f54b5a9ce2a721ffecc3ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/876908691b5743cf9defd1d4ed9a95a0 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fea454f24f44179a0bbe81b677bbd0f : SUCCESS in 12m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d919d80a06248a9b665fa37bb951ca9 : SUCCESS in 13m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a30652b8711d4434a08036d59129636c : SUCCESS in 14m 04s","accounts_in_message":[],"_revision_number":3},{"id":"f34f38902e84d82a863376d3bf8eb1bf7de0eb44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 07:45:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f0ea3c6ab131ddd51ba65accc7b849319b1c2311","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 07:46:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8b32c4d250c4eefa84d8f769b0007a3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74e266e9d2814d98bd59d2002eaa0ef6 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3},{"id":"783b6f30bf387193ae21bbed7430e4ca611c72a2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-13 08:17:05.000000000","message":"Patch Set 3:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":3},{"id":"eff0eb972225079a1a67d8cd0d06f058bd547bf1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-09-15 14:49:11.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ec68e5d147444f3fb9b33b8c8e66c1f5 : SUCCESS in 37m 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e2278994384e3e4e2b36a3e614d0a832a8bd1f50","revisions":{"f061eaeeca8f117cb00326df40e8f8b931826efe":{"kind":"REWORK","_number":1,"created":"2022-08-23 20:10:44.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/66/854266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/66/854266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/854266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/854266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/854266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/66/854266/1"}}},"commit":{"parents":[{"commit":"538e7b2e3321e84c1c814765a2f05de38b098ec5","subject":"Tests: Fix IBM XIV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/538e7b2e3321e84c1c814765a2f05de38b098ec5"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-22 11:08:04.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-23 20:10:20.000000000","tz":120},"subject":"Tests: Fix zone manager tests","message":"Tests: Fix zone manager tests\n\nThere are some zome manager unit tests that will fail depending on the\norder of execution.\n\nThis patch fixes these issues.\n\nChange-Id: I44511900c7e8434398bf9467e5067fc1dfc0940d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f061eaeeca8f117cb00326df40e8f8b931826efe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f061eaeeca8f117cb00326df40e8f8b931826efe"}]},"branch":"refs/heads/master"},"1a556aa19713afa93c9ca6711eaaefe47fb9228e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-08-29 15:06:56.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/66/854266/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/66/854266/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/854266/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/854266/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/854266/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/66/854266/2"}}},"commit":{"parents":[{"commit":"023fa94bde11569a9ba1f68777dba3c3bbaf8d72","subject":"Tests: Fix IBM XIV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/023fa94bde11569a9ba1f68777dba3c3bbaf8d72"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-22 11:08:04.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-24 17:37:38.000000000","tz":120},"subject":"Tests: Fix zone manager tests","message":"Tests: Fix zone manager tests\n\nThere are some zome manager unit tests that will fail depending on the\norder of execution.\n\nThis patch fixes these issues.\n\nChange-Id: I44511900c7e8434398bf9467e5067fc1dfc0940d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1a556aa19713afa93c9ca6711eaaefe47fb9228e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1a556aa19713afa93c9ca6711eaaefe47fb9228e"}]},"branch":"refs/heads/master"},"e2278994384e3e4e2b36a3e614d0a832a8bd1f50":{"kind":"REWORK","_number":3,"created":"2022-08-29 15:10:16.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/66/854266/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/66/854266/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/854266/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/854266/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/66/854266/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/66/854266/3"}}},"commit":{"parents":[{"commit":"023fa94bde11569a9ba1f68777dba3c3bbaf8d72","subject":"Tests: Fix IBM XIV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/023fa94bde11569a9ba1f68777dba3c3bbaf8d72"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-22 11:08:04.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-29 15:10:00.000000000","tz":120},"subject":"Tests: Fix zone manager tests","message":"Tests: Fix zone manager tests\n\nThere are some zome manager unit tests that will fail depending on the\norder of execution.\n\nThis patch fixes these issues.\n\nChange-Id: I44511900c7e8434398bf9467e5067fc1dfc0940d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e2278994384e3e4e2b36a3e614d0a832a8bd1f50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e2278994384e3e4e2b36a3e614d0a832a8bd1f50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
