)]}'
{"id":"openstack%2Fcinder~854269","triplet_id":"openstack%2Fcinder~master~I9325a677376beb44a8c39d9c07d0b48e02a8844f","project":"openstack/cinder","branch":"master","topic":"rbd-native-treads","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2022-09-13 08:20:25.000000000","reason":"Change was submitted"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-09-13 08:20:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9325a677376beb44a8c39d9c07d0b48e02a8844f","subject":"Tests: Fix NFS tests","status":"MERGED","created":"2022-08-23 20:10:44.000000000","updated":"2022-09-15 19:20:46.000000000","submitted":"2022-09-13 08:20:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854269-rbd-native-treads","meta_rev_id":"01d3d6ceb5c6be96cba3bb1628821f5bbd767a7a","_number":854269,"virtual_id_number":854269,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-13 08:20:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-08-29 15:10:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-08-29 15:10:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":1,"date":"2022-08-30 11:29:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-12 18:22:43.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-12 18:22:43.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-23 20:10:59.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-08-23 20:26:52.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2022-08-23 20:34:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-23 23:09:30.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-08-24 03:04:34.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-08-24 03:36:20.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-24 03:59:29.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-08-24 04:08:43.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-24 05:27:35.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-24 07:03:58.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-08-24 10:47:01.000000000","updated_by":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"reviewer":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"state":"REVIEWER"},{"updated":"2022-08-24 13:09:20.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-08-25 05:12:57.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-25 09:15:09.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2022-08-25 10:59:31.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-08-25 14:19:12.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2022-08-25 16:03:14.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2022-08-25 17:47:00.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2022-08-25 21:36:54.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-08-26 03:12:20.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-08-27 11:43:46.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-29 01:38:37.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-30 11:29:42.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-09-01 20:42:11.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-10 19:12:10.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-09-12 18:22:43.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2022-09-15 19:20:46.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"0f8ed4de658e64b82e1f6b560c65705c1ec3dea1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-23 20:10:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfbf3ba2fcdc33b66b755680097a8243da6cda2d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-23 20:10:59.000000000","message":"Patch Set 1:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":1},{"id":"e7059bb8188491cb4ce8ff676f2c4d536e149c37","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-08-23 20:26:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"82094861b03366cae5534cabf3a5b652f97607f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 20:34:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29e2661a82df42cf8f17bbb0951e6a39\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/449825fdb4684e7c97152ddd4231095b : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fb4241de7c74c779b80e46b2a44b5ba : SUCCESS in 11m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bc441f646ff45d88892bbe69efe14ab : SUCCESS in 11m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f172fb946ea640c8b16e9354ad349072 : SUCCESS in 11m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a87b961780674c209e6d277b36b189bb : SUCCESS in 12m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ef280ff725f4480f94df4f05d1617390 : SUCCESS in 15m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2c72cad8e12c4ae2be38a5b93dec680d : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"f15d3f2647de9d573708ca3a9dcad1284aafedc2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 21:14:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc634cf89f0445b29eaf999acc730c64\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0900d90075d4150bcc51f8621f29572 : FAILURE in 21m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8d31b44d2fc49a0b426cfbaf315fd5f : SUCCESS in 49m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"290cab85e8bbaea4c2361a5ad93f1bcc994f40e6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-23 23:09:30.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661285456421 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"6483567ab8fc8987d54ffe8a32b23b84cc7450e3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-24 03:04:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/69/854269/1/check/dsvm-tempest-cisco-zonemanager-job/b5bc99a/ : POST_FAILURE in 12m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a4d186da5af6f426303ef0d6dcca0de453d0252a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-24 03:36:20.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-69-854269-1/ns5_nfs : SUCCESS in 3h 38m 53s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-69-854269-1/ns5_iscsi : SUCCESS in 3h 39m 22s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"b2ad3affc36e5304d44071015da4f44821f7f6c7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-24 03:59:29.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/69/854269/1/check/iscsi-driver-alletra-6k/f9817ab/f9817ab.html : FAILURE in 1h 06m 25s","accounts_in_message":[],"_revision_number":1},{"id":"76c38a7ca63ce8aa8acfd059dadd5aca98d4cb69","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-24 04:08:43.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/854269 : SUCCESS in 1h 23m 53s","accounts_in_message":[],"_revision_number":1},{"id":"f335d3f36fa5cfa8f466094e54f6f8ef3d8a112e","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-24 05:27:35.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/69/854269/1/check/fujitsu-eternus-dx-iscsi/5f104cf : SUCCESS in 1h 39m 51s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/69/854269/1/check/fujitsu-eternus-dx-fc/47d1e13 : SUCCESS in 1h 46m 31s","accounts_in_message":[],"_revision_number":1},{"id":"781f6ef16ed5f5bab00f5eb1e0038b912132f238","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-24 07:03:58.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/854269/1/iscsi/ : FAILURE in 1h 41m 36s\n- cinder-smb http://cloudbase-ci.com/cinder/854269/1/smb/ : SUCCESS in 53m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ed2cd34fb1f38d9f7c9f5c5576f477c4bad3b7cc","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-24 08:27:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f0a448ac2ba98e955d71965632767a47ed333b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-24 08:41:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"89bf027c0b891d4f9020b2bf27415e011dbf46f6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-24 08:41:15.000000000","message":"Patch Set 2:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2},{"id":"f44a10d26c5857970081262f12e4e73cf698f42b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 09:06:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1d2a617f36c48669404848b622881a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a29d0b13a9924d7ca8daf16fd9cce5aa : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b830dd568474d778476da62960029c9 : SUCCESS in 11m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efd25583c329402ca9af364168456b5a : SUCCESS in 12m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/718e4df36925481da9bbea444d382843 : SUCCESS in 11m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64b15dca50e74ef18c0fddae0005229e : SUCCESS in 13m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/15f93be15a3b4095b150ec7af994aa48 : SUCCESS in 16m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/25abcf4ccb794d9fbd76dbda0a7a73bd : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":2},{"id":"bde7537ed4e4fe01fcea6116b9f82af9143b5b6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 09:12:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e95cb0a25e704f0fa7b254063f321338\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11ee5078f0d44a6d987ab8e00fcb8842 : SUCCESS in 19m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a3afd77975c44e380c55d0ad125d407 : SUCCESS in 24m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7b7da9e8e7100ab84a8cebe2264fe8e96ee02eb2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-24 09:24:10.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/69/854269/2/check/iscsi-driver-alletra-6k/b0b2afb/b0b2afb.html : FAILURE in 31m 52s","accounts_in_message":[],"_revision_number":2},{"id":"828c86ebdd9af2d82240a3e105e54031358541ba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-24 09:26:11.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661330484180 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"8dba4b5c1681a5bf455368f76a20376e4814701b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-24 09:49:50.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/854269 : FAILURE in 1h 02m 50s","accounts_in_message":[],"_revision_number":2},{"id":"54aa5a43405e41ca95921e2e187c23566132ef21","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-24 10:26:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/854269/2/iscsi/ : SUCCESS in 1h 17m 16s\n- cinder-smb http://cloudbase-ci.com/cinder/854269/2/smb/ : SUCCESS in 1h 09m 08s","accounts_in_message":[],"_revision_number":2},{"id":"79b84f6d2316dc5cf9e05fa58f719f535f8a5f19","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2022-08-24 10:47:01.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7711aff88065d0f165d3ecdcd103e3e3befb5644","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-24 11:50:32.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/69/854269/2/check/dsvm-tempest-cisco-zonemanager-job/15f431e/ : POST_FAILURE in 12m 31s","accounts_in_message":[],"_revision_number":2},{"id":"03c6b7195b162ff519d29aa7fa1cc4ea8a0ce6c2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-24 11:58:02.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/854269 : SUCCESS in 1h 29m 05s","accounts_in_message":[],"_revision_number":2},{"id":"3c314fbbf39bd60a31f92a2a61741621c06e333f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-24 12:16:27.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-69-854269-2/ns5_nfs : SUCCESS in 3h 32m 39s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-69-854269-2/ns5_iscsi : SUCCESS in 3h 34m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"7379520bcf9ba999cd8ebbeb78f2c5532dac9be5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-24 13:09:20.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/24_08_2022_12_46_54\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"38654a3744ae7ab5f7ddbdb124e62c9902f53ed4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-08-24 18:58:50.000000000","message":"Patch Set 2:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/24_08_2022_18_42_13\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"5b03aa0060825de7c6f48c236fffa71cc5868caa","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-25 05:12:57.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/69/854269/2/check/DellEMC_PowerFlex/68f7000/DellEMC_PowerFlex/1211 : ABORTED in 2h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"e481ec2dd9e4ec7c0c8601ea1b4271eec4cf8cf3","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-08-25 09:15:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/854269-1/HPMSA-iSCSI/ : SUCCESS in 56:28.14\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/854269-1/HPMSA-FC/ : SUCCESS in 54:53.74","accounts_in_message":[],"_revision_number":1},{"id":"bd1fb4e248ebd83a082e6abc0ac263728a2f8813","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-25 10:59:31.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/69/854269/2/check/DellEMC_PowerStore_FC/d3ea19f/DellEMC_PowerStore_FC/779 : SUCCESS in 1h 16m 05s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/69/854269/2/check/DellEMC_PowerStore_NFS/1015286/DellEMC_PowerStore_NFS/319 : SUCCESS in 1h 29m 30s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/69/854269/2/check/DellEMC_PowerStore_NVMeTCP/3930d26/DellEMC_PowerStore_NVMeTCP/194 : SUCCESS in 1h 45m 55s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"0c81740d6607ff1e5a5fdf5ae6dfd91e6c98e4f1","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-08-25 14:19:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/854269-1/PVME-iSCSI/ : SUCCESS in 1:08:10\n- PVME-FC https://os-logs.seagate.com/dell-ci/854269-1/PVME-FC/ : SUCCESS in 1:08:09","accounts_in_message":[],"_revision_number":1},{"id":"1d22ecbb73a9030f4b2d39948dda33a77ebec9eb","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-08-25 16:03:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/854269-1/Lenovo-iSCSI/ : SUCCESS in 1:12:55\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/854269-1/Lenovo-FC/ : SUCCESS in 1:26:51","accounts_in_message":[],"_revision_number":1},{"id":"7b1c98036f74afb6fab1d08354baf527832e7d4f","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-08-25 17:47:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/854269-1/Seagate-iSCSI/ : SUCCESS in 1:13:31\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/854269-1/Seagate-FC/ : SUCCESS in 1:13:04","accounts_in_message":[],"_revision_number":1},{"id":"422399d0f15412dd87ba2a758a84f00abc3a1e38","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-25 21:36:54.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/69/854269/2/check/DellEMC_Unity_FC/e0ccc31/DellEMC_Unity_FC/1548 : FAILURE in 14m 14s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/69/854269/2/check/DellEMC_Unity_iSCSI/a1d5180/DellEMC_Unity_iSCSI/1426 : SUCCESS in 1h 48m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"046f08587861b69dbd6a88546992a9ceae082f88","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-08-26 01:43:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/854269-2/HPMSA-iSCSI/ : SUCCESS in 59:34.33\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/854269-2/HPMSA-FC/ : SUCCESS in 59:34.45","accounts_in_message":[],"_revision_number":2},{"id":"8b023e940f52dcd38a8d2432ebb56b384503d91c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-08-26 03:12:20.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"99d59bb5e8bbfa53d105258c75e9f95570273f77","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-08-26 09:19:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/854269-2/PVME-iSCSI/ : SUCCESS in 1:07:23\n- PVME-FC https://os-logs.seagate.com/dell-ci/854269-2/PVME-FC/ : SUCCESS in 1:07:16","accounts_in_message":[],"_revision_number":2},{"id":"f509835eff601465d77e2c5a0aa900677b152a0f","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-08-26 11:22:41.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/854269-2/Lenovo-iSCSI/ : SUCCESS in 1:05:53\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/854269-2/Lenovo-FC/ : SUCCESS in 1:03:59","accounts_in_message":[],"_revision_number":2},{"id":"1f29a3dc1b6b051b9105c7a6ebcb29a6ee0de60e","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-08-26 14:56:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/854269-2/Seagate-iSCSI/ : SUCCESS in 1:17:48\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/854269-2/Seagate-FC/ : SUCCESS in 1:15:28","accounts_in_message":[],"_revision_number":2},{"id":"f36347e2c075bd9b99fddcc130299f98cddcb176","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-08-27 11:43:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/69/854269/2/20220827171315/NEC-ISCSI : SUCCESS in 3h 28m 47s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/69/854269/2/20220827171315/NEC-FC : SUCCESS in 3h 30m 28s","accounts_in_message":[],"_revision_number":2},{"id":"26fe9ef4fcb4ec714e50a458ef021dbb86840086","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-08-29 01:38:37.000000000","message":"Patch Set 1:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":1},{"id":"5ce283fafdd00fc9b281fbd22b679b011e54738c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-08-29 01:39:11.000000000","message":"Patch Set 2:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":2},{"id":"ce0871a55bc89cc440353c26ea8c3762b28e0c54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-29 15:06:56.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"b9e3857ee2291f185c8cf7952a9220e6d27a2afe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-29 15:07:09.000000000","message":"Patch Set 3:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":3},{"id":"a6c62147207d0b7ff8089a4659743425fda7b15b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-08-29 15:10:16.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"a25584174757cc6882986bc17d7ad8b729061ab9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-08-29 15:10:24.000000000","message":"Patch Set 4:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":4},{"id":"d371189b69d2d82693d0c9fa287b4cdddaaa0556","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-29 15:33:19.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661785628424 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"82d4dd6fe23b9793f43c63e252179e81a66a5197","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 15:38:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cc6389350cc4305ab0fecdefd1caf4f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8880271cff704106b618ad397c28ab8f : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e5b255de58045539e1cbc6068f8ef1d : SUCCESS in 12m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5be920d7f1a4718860b88b4b03ccefc : SUCCESS in 13m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd4a90dd94224daca6cc53ecbdd5f264 : SUCCESS in 13m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/194898f2692040b39c5c516c111f443a : SUCCESS in 15m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b02f260340ef43339967a99dbe7d0437 : SUCCESS in 16m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c75b513ca0034031bd427b1fc1c2ab4d : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":4},{"id":"09250e504ac051a6f2719d48094bf0ecc5a69934","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 16:07:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1caafc401244e88a0e5d70b3dd32e83\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9eebfc85a56943ce90250978a6d2825a : SUCCESS in 25m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/08539b79d6a746108a7a132ca03dffd7 : FAILURE in 27m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"014776eaa12c1ce1fbb18d3f1ba578550c70aacb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-29 20:53:39.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/69/854269/4/check/iscsi-driver-alletra-6k/8b4d41a/8b4d41a.html : FAILURE in 1h 38m 52s","accounts_in_message":[],"_revision_number":4},{"id":"e7db5ed06c20778e475def48cc9c70304b0a642a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-29 21:44:47.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-69-854269-3/ns5_nfs : SUCCESS in 3h 27m 38s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-69-854269-3/ns5_iscsi : SUCCESS in 3h 36m 54s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"5ad307500b3372a31fab08a5608dd92af089c385","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-29 21:54:47.000000000","message":"Patch Set 4:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/69/854269/4/check/fujitsu-eternus-dx-iscsi/c943daf : SUCCESS in 1h 40m 24s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/69/854269/4/check/fujitsu-eternus-dx-fc/c646d5e : SUCCESS in 1h 39m 26s","accounts_in_message":[],"_revision_number":4},{"id":"9b286347e1c23865f4918fa9599ec6e3b9da85d9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-29 22:03:44.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/854269 : FAILURE in 1h 34m 14s","accounts_in_message":[],"_revision_number":4},{"id":"217d3c80021aba19d12ec4df343c5d60b22dceb6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-29 22:29:08.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/854269 : FAILURE in 1h 32m 41s","accounts_in_message":[],"_revision_number":4},{"id":"581c0937ff91c37ab6e2678c791f07596c8f03c3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-29 23:02:47.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/854269/4/iscsi/ : SUCCESS in 1h 15m 20s\n- cinder-smb http://cloudbase-ci.com/cinder/854269/4/smb/ : SUCCESS in 1h 08m 59s","accounts_in_message":[],"_revision_number":4},{"id":"d7aa1e1ff6cad60ff496d95257ef2eb2054eb5d1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-30 02:58:05.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661785828437 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"731a90ae7ffff95dac99f02d2fc6c80327d89b59","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-30 03:47:35.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-69-854269-4/ns5_nfs : SUCCESS in 3h 37m 53s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-69-854269-4/ns5_iscsi : SUCCESS in 3h 39m 24s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"0ce093277fd551dd25c7af4a72d2ce1c89e6b6ec","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-08-30 11:29:42.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e8264a8a353b74cf2eeed37fff9f5e7ad84fc5b1","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-08-30 16:37:50.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/69/854269/4/check/DellEMC_PowerStore_FC/17079f3/DellEMC_PowerStore_FC/816 : SUCCESS in 1h 29m 37s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/69/854269/4/check/DellEMC_PowerStore_NFS/62a6609/DellEMC_PowerStore_NFS/354 : SUCCESS in 1h 34m 02s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/69/854269/4/check/DellEMC_PowerStore_NVMeTCP/277444f/DellEMC_PowerStore_NVMeTCP/229 : SUCCESS in 1h 47m 05s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"c094f4f2f9855fc23aa572ab7dd39787685b7e36","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-08-30 17:05:43.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/69/854269/4/check/DellEMC_PowerFlex/04f68cd/DellEMC_PowerFlex/1246 : ABORTED in 2h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"42aa695e1f29f031bb33c17e7e80014c9950abc6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-30 18:12:40.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/69/854269/4/check/DellEMC_Unity_FC/0e66316/DellEMC_Unity_FC/1605 : FAILURE in 14m 46s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/69/854269/4/check/DellEMC_Unity_iSCSI/48ab6d3/DellEMC_Unity_iSCSI/1472 : FAILURE in 2h 10m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"4e16f468a498f7abbb1fc91bf92bf495f22bd896","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2022-08-30 23:36:25.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/854269-4/HPMSA-iSCSI/ : SUCCESS in 57:43.84\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/854269-4/HPMSA-FC/ : SUCCESS in 54:55.25","accounts_in_message":[],"_revision_number":4},{"id":"03a8ee230144502582b82a231f0b411063897c7d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-08-31 00:24:40.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/69/854269/4/check/dsvm-tempest-cisco-zonemanager-job/bb1e1bf/ : FAILURE in 1h 03m 52s","accounts_in_message":[],"_revision_number":4},{"id":"edba313a59ff7c489574890b6050b551c3f26303","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2022-08-31 04:18:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/854269-4/PVME-iSCSI/ : SUCCESS in 1:04:43\n- PVME-FC https://os-logs.seagate.com/dell-ci/854269-4/PVME-FC/ : SUCCESS in 1:05:36","accounts_in_message":[],"_revision_number":4},{"id":"099b64b1a13cf20921b47ffa4e6b77516d8db518","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2022-08-31 05:31:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/854269-4/Lenovo-iSCSI/ : SUCCESS in 1:04:45\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/854269-4/Lenovo-FC/ : SUCCESS in 1:02:51","accounts_in_message":[],"_revision_number":4},{"id":"66050c3b8c050627cdca946ba793caa62dd53341","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2022-08-31 08:30:42.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/854269-4/Seagate-iSCSI/ : SUCCESS in 1:13:20\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/854269-4/Seagate-FC/ : SUCCESS in 1:12:06","accounts_in_message":[],"_revision_number":4},{"id":"5ac5961f249a0f4886d8e890545d083edbd8b7ee","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-09-01 20:42:11.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/69/854269/4/ : SUCCESS in 1h 19m 45s","accounts_in_message":[],"_revision_number":4},{"id":"0c1acf37496a6f96fb57ba298f536d5a0a94f3cf","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-09-03 14:39:13.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/69/854269/4/20220903165219/NEC-ISCSI : SUCCESS in 3h 28m 04s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/69/854269/4/20220903165219/NEC-FC : SUCCESS in 6h 46m 51s","accounts_in_message":[],"_revision_number":4},{"id":"3096fc1567f03f43f55d6d7c119a7a1cd4cce132","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-09-10 19:12:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/69/854269/4/20220910205052/NEC-V-FC : SUCCESS in 5h 56m 41s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/69/854269/4/20220910205052/NEC-V-ISCSI : SUCCESS in 7h 21m 15s","accounts_in_message":[],"_revision_number":4},{"id":"2dd9d8d7dbe6efa0b88a75b7916ddb550bd6f8cd","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-09-12 18:22:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"369cb290e62cd2b09c3030d2b72f8a0b1a5a8c3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 08:00:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"79234578046651ae0e72a1abc4d1c94fb46a3437","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 08:20:25.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6fca8932b1b4e5c996bde4977d587cc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f6a38c2a73a402eba67ac440a01317f : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00f584df61444c43870ecca7dcc07bc0 : SUCCESS in 11m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70840c5d8b77418e944bc120f0b5919d : SUCCESS in 11m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3350b7a93e08453880a20ebe0265c86c : SUCCESS in 18m 21s","accounts_in_message":[],"_revision_number":4},{"id":"1eb25a98980524bec2f34aa165134a619af348a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 08:20:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"cea6da1e9af5bba62b2d3260194c6d1acb85ad49","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 08:21:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7251de8d00b6428bb737da9712c8a1cb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/125c8b53879e46ebac0e9bcf3cef0f88 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":4},{"id":"dca7cdef1324d5f661c69e9aee88b34b5cfe5cce","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-13 08:24:27.000000000","message":"Patch Set 4:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":4},{"id":"01d3d6ceb5c6be96cba3bb1628821f5bbd767a7a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-09-15 19:20:46.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3b90a9ab6f7543588eb2599dce0cdd9b : TIMED_OUT in 3h 05m 33s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"685e35c9b8e0cee93ebeb9ba938bbd9ee865144e","revisions":{"5c4138d3a3f360ea5f4b46a116e121864b87cd83":{"kind":"REWORK","_number":1,"created":"2022-08-23 20:10:44.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/69/854269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/854269/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/854269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/854269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/854269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/854269/1"}}},"commit":{"parents":[{"commit":"d67d86a0bf2b909dd6df36b831332a547f28ca06","subject":"Tests: Fix NetApp tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d67d86a0bf2b909dd6df36b831332a547f28ca06"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-23 14:27:15.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-23 20:10:20.000000000","tz":120},"subject":"Tests: Fix NFS tests","message":"Tests: Fix NFS tests\n\nThere are 2 NFS unit tests that won\u0027t run successfully on their own and\ndepend on other tests from the same file to be run previously to work as\nintended:\n\n- test_create_sparsed_volume\n- test_create_nonsparsed_volume\n\nThe issue is caused by calling the self.override_config method when we\nare actually using a Mock in the driver for the configuration.\n\nThis patch replaces the calls with simple assignments and adds a comment\nexplaining why this is the right way to do it.\n\nWe only make these changes in the NfsDriverTestCase class and not the\nNfsDriverDoSetupTestCase class because the latter uses a real class and\ncan use the override mechanism.\n\nChange-Id: I9325a677376beb44a8c39d9c07d0b48e02a8844f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c4138d3a3f360ea5f4b46a116e121864b87cd83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5c4138d3a3f360ea5f4b46a116e121864b87cd83"}]},"branch":"refs/heads/master"},"eef26ba18eae0ddae0d6dee3bf1a63250da5a2ab":{"kind":"REWORK","_number":2,"created":"2022-08-24 08:41:12.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/69/854269/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/854269/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/854269/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/854269/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/854269/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/854269/2"}}},"commit":{"parents":[{"commit":"d67d86a0bf2b909dd6df36b831332a547f28ca06","subject":"Tests: Fix NetApp tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d67d86a0bf2b909dd6df36b831332a547f28ca06"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-23 14:27:15.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-24 08:35:21.000000000","tz":120},"subject":"Tests: Fix NFS tests","message":"Tests: Fix NFS tests\n\nThe NFS and Remotefs unit tests use a mocking of the cinder\nConfiguration object as the driver configuration.\n\nThis makes the usage of self.override_config not work as expected, and a\ncouple of tests actually use it, so it will only work if the test run\nhas run a test that forcefully sets the value.\n\nIf we have these tests:\n\n  def test1(self):\n      self.configuration.option \u003d True\n      ...\n\n  def test2(self):\n      self.set_override(\u0027option\u0027, False)\n      ...\n\nIt will fail if test2 runs before test1.\n\nThere are currently 2 tests that use the override_config in the nfs UT\nfile and where its success depend on the test runnin order:\n\n- test_create_sparsed_volume\n- test_create_nonsparsed_volume\n\nThis patch removes the mocking and uses a proper Configuration so that\nthe whole file can use the standard override_config method.\n\nChange-Id: I9325a677376beb44a8c39d9c07d0b48e02a8844f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eef26ba18eae0ddae0d6dee3bf1a63250da5a2ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eef26ba18eae0ddae0d6dee3bf1a63250da5a2ab"}]},"branch":"refs/heads/master"},"d3cbad37d06990c3a078e8027421cdcf42200d39":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-08-29 15:06:56.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/69/854269/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/854269/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/854269/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/854269/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/854269/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/854269/3"}}},"commit":{"parents":[{"commit":"d2802729a43bbde57175786a69f262a2e559f0fe","subject":"Tests: Fix NetApp tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d2802729a43bbde57175786a69f262a2e559f0fe"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-23 14:27:15.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-24 17:37:38.000000000","tz":120},"subject":"Tests: Fix NFS tests","message":"Tests: Fix NFS tests\n\nThe NFS and Remotefs unit tests use a mocking of the cinder\nConfiguration object as the driver configuration.\n\nThis makes the usage of self.override_config not work as expected, and a\ncouple of tests actually use it, so it will only work if the test run\nhas run a test that forcefully sets the value.\n\nIf we have these tests:\n\n  def test1(self):\n      self.configuration.option \u003d True\n      ...\n\n  def test2(self):\n      self.set_override(\u0027option\u0027, False)\n      ...\n\nIt will fail if test2 runs before test1.\n\nThere are currently 2 tests that use the override_config in the nfs UT\nfile and where its success depend on the test runnin order:\n\n- test_create_sparsed_volume\n- test_create_nonsparsed_volume\n\nThis patch removes the mocking and uses a proper Configuration so that\nthe whole file can use the standard override_config method.\n\nChange-Id: I9325a677376beb44a8c39d9c07d0b48e02a8844f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d3cbad37d06990c3a078e8027421cdcf42200d39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d3cbad37d06990c3a078e8027421cdcf42200d39"}]},"branch":"refs/heads/master"},"685e35c9b8e0cee93ebeb9ba938bbd9ee865144e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-08-29 15:10:16.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/69/854269/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/69/854269/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/854269/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/854269/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/69/854269/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/69/854269/4"}}},"commit":{"parents":[{"commit":"8524779a6a02be6b83900fce7299473bbec5d14d","subject":"Tests: Fix NetApp tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8524779a6a02be6b83900fce7299473bbec5d14d"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-23 14:27:15.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-08-29 15:10:02.000000000","tz":120},"subject":"Tests: Fix NFS tests","message":"Tests: Fix NFS tests\n\nThe NFS and Remotefs unit tests use a mocking of the cinder\nConfiguration object as the driver configuration.\n\nThis makes the usage of self.override_config not work as expected, and a\ncouple of tests actually use it, so it will only work if the test run\nhas run a test that forcefully sets the value.\n\nIf we have these tests:\n\n  def test1(self):\n      self.configuration.option \u003d True\n      ...\n\n  def test2(self):\n      self.set_override(\u0027option\u0027, False)\n      ...\n\nIt will fail if test2 runs before test1.\n\nThere are currently 2 tests that use the override_config in the nfs UT\nfile and where its success depend on the test runnin order:\n\n- test_create_sparsed_volume\n- test_create_nonsparsed_volume\n\nThis patch removes the mocking and uses a proper Configuration so that\nthe whole file can use the standard override_config method.\n\nChange-Id: I9325a677376beb44a8c39d9c07d0b48e02a8844f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/685e35c9b8e0cee93ebeb9ba938bbd9ee865144e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/685e35c9b8e0cee93ebeb9ba938bbd9ee865144e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
