)]}'
{"id":"openstack%2Fcinder~854954","triplet_id":"openstack%2Fcinder~master~Ia36b8892e88453716f68956c44daeb14d8854912","project":"openstack/cinder","branch":"master","topic":"fujitsu-driver-update","attention_set":{},"removed_from_attention_set":{"33609":{"account":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"last_update":"2023-09-15 01:16:38.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia36b8892e88453716f68956c44daeb14d8854912","subject":"Fujitsu Driver: Add Timeout function","status":"ABANDONED","created":"2022-08-29 08:04:33.000000000","updated":"2023-09-15 04:00:07.000000000","total_comment_count":12,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"77451a749957f2da3f4ab1c3f4fac567bb529b6b","_number":854954,"virtual_id_number":854954,"owner":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-23 05:16:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-29 08:28:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-29 08:28:53.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-08-29 10:29:41.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-29 10:34:31.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-08-29 10:37:51.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-29 10:58:22.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-08-29 11:03:27.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-08-29 11:13:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-29 11:55:36.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-29 15:36:12.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-09-01 13:33:56.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-01 20:35:56.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-09-07 09:54:24.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-07 17:59:03.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-09-15 06:52:21.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-11 03:58:28.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2022-10-11 08:26:18.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-10-11 08:36:36.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-10-11 09:02:47.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-11 09:12:13.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-10-11 09:31:07.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-10-11 09:31:51.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-10-11 09:53:02.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-11-21 06:05:26.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-22 03:25:51.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2023-04-06 10:14:19.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-05-23 05:18:05.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2023-05-23 05:23:19.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-05-23 05:41:14.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-05-23 05:41:24.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-09-15 04:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"}],"messages":[{"id":"af41abfb587d3fb1f61cc4d16745116520dc5883","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-08-29 08:04:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d8789ea391f8b8d8b79ca86065a7318303a4479","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 08:28:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9197c64859304709b7ac9007f0c02ca8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c8e57b6858a47e4872777bf4cf7c6dc : SUCCESS in 21m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c520d2a9621b4779abcd17b8537da3c3 : SUCCESS in 15m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5fe1d0b5183d67529dfe404d7daf86e3c478f75a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-29 08:28:53.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661760282620 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"6b0e49e8c342cfc2e76c3afe86d0d535b9376363","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-08-29 08:55:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fece4dc991913f442e54a4e71e1b3a236c2d1d8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-08-29 08:58:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"79a96f2c94b24fbe27351925106abc654b1f101c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-29 09:17:52.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661763312704 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"8284db039e194e6394542d52129d17b02ca86791","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 09:28:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3219a047fb54668b6d712f3e4dadd6e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd08a7b1afdd406bb565f574474e2e68 : FAILURE in 23m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0bb6d23e2e8440fd82cb844104bdb837 : SUCCESS in 13m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5e5b9b86db9278e365a98981f9bf4b923732be16","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-08-29 09:40:45.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1661763507703 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"f92895b4432b8a169a315358fddeb1efbf2158ce","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-29 10:29:41.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/854954 : SUCCESS in 1h 26m 17s","accounts_in_message":[],"_revision_number":3},{"id":"e9379798663e3d047cc2fe1d45d666cb59a3eea7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-08-29 10:34:31.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/54/854954/3/check/iscsi-driver-alletra-6k/e3c09c9/e3c09c9.html : FAILURE in 1h 35m 00s","accounts_in_message":[],"_revision_number":3},{"id":"1f0497af15d0b099678141b2c71e61b9bbd5757b","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-08-29 10:37:51.000000000","message":"Patch Set 3:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/54/854954/3/check/fujitsu-eternus-dx-iscsi/068d79e : SUCCESS in 1h 27m 41s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/54/854954/3/check/fujitsu-eternus-dx-fc/405f0cc : SUCCESS in 1h 38m 55s","accounts_in_message":[],"_revision_number":3},{"id":"d16a6905cf81a63029648ba6c5f793b7d64bb76b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-08-29 10:43:29.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/854954 : SUCCESS in 1h 41m 00s","accounts_in_message":[],"_revision_number":3},{"id":"da28e8f14fb4f67fb148312abeebd1eb81508aa1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-29 10:58:22.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/54/854954/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5d14ef5/ : SUCCESS in 1h 56m 06s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/54/854954/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/5ab3155/ : SUCCESS in 1h 30m 20s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/54/854954/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/b8da8b6/ : SUCCESS in 1h 52m 31s","accounts_in_message":[],"_revision_number":3},{"id":"1150509a797f60d5147847275b60a6057ec974d4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-29 11:03:27.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/54/854954/3/check/DellEMC_Unity_FC/826e4e9/DellEMC_Unity_FC/1591 : FAILURE in 12m 56s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/54/854954/3/check/DellEMC_Unity_iSCSI/5e0cd6c/DellEMC_Unity_iSCSI/1458 : SUCCESS in 2h 04m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"c96e4a53b07a195db46b8af11300f2f17e4f7d76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 11:13:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bbc25f2f08b74a46b4d2ea45c4da9fcc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49a9f228d91d47c99450fc19aca59e49 : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08f9d466b15140a19c80e487aee6cb95 : FAILURE in 20m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccfcbd16f9d143a5a062674c41b1e6e8 : SUCCESS in 11m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d85b61df2cae429f8ba50e1037ee9bde : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/650adf72ca904e959de1f70410d597f3 : SUCCESS in 13m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/3a0ce71f60534645ae5687bee6b62a6b : SUCCESS in 1h 02m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/140d3ead37f44f1a8af21e8cc5c9a0e5 : SUCCESS in 1h 30m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7c00331895e4b23ba4738904c586bc7 : SUCCESS in 47m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3e4833799ff64066b22fe4e33a6e387c : SUCCESS in 15m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/27bcc3797de04d798e9ee905c21715ae : SUCCESS in 4m 27s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7d598de78398420298d03254deaa4868 : FAILURE in 6m 46s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/57e17ba10c574622b1dfd612935fe2a7 : SUCCESS in 6m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a6f1a5d624bc460c80786cf1b6418145 : SUCCESS in 7m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3a34ce7d6699401d949477c1820be8b6 : SUCCESS in 39m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/627d60f3fa71475fb6ffa2d059c96cd7 : FAILURE in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/95f88ab3afc84debbbe53d47327dd1e8 : SUCCESS in 1h 23m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e7b46138626d458482dfb03c0cf539d3 : FAILURE in 1h 22m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f51dcc7d77614bf29fd753d5a75cdbb8 : SUCCESS in 1h 38m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2c63769e5cf44e4581f1729f96ff7f09 : SUCCESS in 56m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cdec9f352ba74e0d9de8fea63841c80b : SUCCESS in 2h 07m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/190b40656bd54656b3b5a9e6af300618 : SUCCESS in 34m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/600102b4c8fe4d6da040df461152e5f8 : SUCCESS in 1h 26m 42s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6a9ddba892914f0c9120fa0356b7f273 : SUCCESS in 2h 00m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90200a263620442caf36057b0aec4d0f : SUCCESS in 50m 49s","accounts_in_message":[],"_revision_number":3},{"id":"f914fa061cee6b67cc98ec9833dac7835b0af94c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-29 11:55:36.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/54/854954/3/check/tempest-dsvm-full-focal-py3/ecea498/ : SUCCESS in 1h 38m 34s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/54/854954/3/check/devstack-plugin-ceph-dsvm-focal-py3/c62e9b8/ : SUCCESS in 1h 47m 10s","accounts_in_message":[],"_revision_number":3},{"id":"de8dd35c689b32a72f3c1f952288f24f3b9f1533","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-29 15:36:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/854954/3/iscsi/ : SUCCESS in 1h 04m 57s\n- cinder-smb http://cloudbase-ci.com/cinder/854954/3/smb/ : SUCCESS in 1h 06m 12s","accounts_in_message":[],"_revision_number":3},{"id":"db2b1dcbcbf2bac3fff0774c7dd7c0303b56585f","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-08-30 01:16:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bcec2538385c4ef1fed70328449e6ab6538a1432","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 01:30:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2fc1984cd154345955685068cbb0dd9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/abce40a9497241e799b4760a3936804a : SUCCESS in 12m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fdbcf1a4a0754968bf67ed62ac510c76 : SUCCESS in 12m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5261c619e4967b2ec0af51c81c366995dd12cf01","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-30 03:13:53.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/54/854954/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/209c31a/ : SUCCESS in 1h 55m 45s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/54/854954/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/87e456a/ : SUCCESS in 1h 32m 15s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/54/854954/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/71a754e/ : SUCCESS in 1h 49m 49s","accounts_in_message":[],"_revision_number":3},{"id":"dd3da0d9c5bfafc7ba2430831191a761180aeccb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 03:19:16.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/71db6a994dff4507a93dc27a174060f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c783eb10011475397a1a3559ead759a : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f8d7bd71901429489c1e8b1bc979e58 : TIMED_OUT in 40m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00c9bc732d574a7fa570d032255caa96 : SUCCESS in 11m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98f0b9a03d6f4e908feccc23653fea66 : SUCCESS in 10m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/094c980dd13d4d3cb6e8c28edc19202e : SUCCESS in 10m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/567679f21a874ea1aa12251e516a61c9 : SUCCESS in 1h 09m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cd0af7f192c54ee191ffc2326b88caba : SUCCESS in 1h 36m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1c5f1a7a161f4b779424a62a62600198 : SUCCESS in 52m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c9269cd924dc4160b394476cbe86ba23 : TIMED_OUT in 41m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9574aace022b45d7962a102d50b6d7d4 : SUCCESS in 4m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/973a002e7799443ba0a09b15bab86450 : FAILURE in 7m 08s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/191dae6885de4c6997f7bc6a084fa4ef : SUCCESS in 5m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/38f0ee7ca78b44bfb221e8a87506cfe4 : SUCCESS in 6m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/081712f48e9a4391831761fdbe7eed74 : SUCCESS in 43m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d7e6263b158f4fa886ab0137f6563b5c : FAILURE in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/47db613602194f8d9dfc289b65d291f4 : SUCCESS in 1h 23m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3bc6bc92700d41bd90ed265b945b8cc2 : FAILURE in 1h 37m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b93f1dd596d94f14b16334e7dd7241ef : SUCCESS in 1h 46m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/547455795b3c41a4ac84816e05d7d6f9 : SUCCESS in 1h 08m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/462e7e1d5e1d447789ce446eee2a0840 : SUCCESS in 1h 27m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/408498c4304d4ffcb475c68868f3201a : SUCCESS in 41m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9bd630f74eb24ec1ae49d11588dfb8e5 : SUCCESS in 1h 34m 04s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b9204f8ba7d94eb99af26482dc820791 : SUCCESS in 2h 00m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4dd7c06580c84935a816ca3d490c80d9 : SUCCESS in 53m 08s","accounts_in_message":[],"_revision_number":3},{"id":"5412f8420007d0b717486c35d762f7dcfd1c62c3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-30 03:31:36.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/54/854954/3/check/tempest-dsvm-full-focal-py3/6e9df9e/ : SUCCESS in 1h 26m 22s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/54/854954/3/check/devstack-plugin-ceph-dsvm-focal-py3/de0177f/ : SUCCESS in 1h 38m 22s","accounts_in_message":[],"_revision_number":3},{"id":"acda753dbac6df2f6935d59934da8174a47f0b75","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-08-30 04:55:53.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/854954/3/iscsi/ : SUCCESS in 1h 21m 23s\n- cinder-smb http://cloudbase-ci.com/cinder/854954/3/smb/ : SUCCESS in 1h 20m 51s","accounts_in_message":[],"_revision_number":3},{"id":"a86108bcee307fe1927bd38151d171c4ca746a6f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-09-01 13:33:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/854954/3/20220901152209/NEC-ISCSI : SUCCESS in 7h 10m 02s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/854954/3/20220901152209/NEC-FC : SUCCESS in 7h 11m 44s","accounts_in_message":[],"_revision_number":3},{"id":"a86878eea76a9a39a59cc24c9ec4440fdc0fbab4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-09-01 20:35:56.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/54/854954/3/check/dsvm-tempest-cisco-zonemanager-job/60a1131/ : FAILURE in 1h 08m 27s","accounts_in_message":[],"_revision_number":3},{"id":"b54f3e707230226beef4cc2fae91045718229508","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-09-07 09:54:24.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/54/854954/3/ : SUCCESS in 1h 22m 37s","accounts_in_message":[],"_revision_number":3},{"id":"b9e26813696095cc6f38de7d41db07f0b47dd771","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-09-07 17:59:03.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/854954/3/20220907210638/NEC-V-ISCSI : SUCCESS in 2h 23m 41s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/854954/3/20220907210638/NEC-V-FC : SUCCESS in 5h 52m 22s","accounts_in_message":[],"_revision_number":3},{"id":"cf02c65e22d42fafbb9bfa6320d949180cb975f3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-09-15 06:52:21.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6a7cb8c397794df0b42bb70e9a3e4ee6 : SUCCESS in 40m 40s","accounts_in_message":[],"_revision_number":3},{"id":"d16e9b25271db8f8e2e26c824526bf864da4fe31","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-10-11 02:56:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5ff7b66d2f96855c0aa763366fb75bb2030275d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 03:28:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c9378f98ccd45c983ce5d6262545c44\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f7af30c0d9364fef83d3ee783b14b484 : SUCCESS in 28m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9c0c6ca04b5145e3adc439f517332cf0 : SUCCESS in 23m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0931d6c97480e9d8a72f882de654e5f68db31c82","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-10-11 03:49:23.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/854954/3/iscsi/ : FAILURE in 52m 31s\n- cinder-smb http://cloudbase-ci.com/cinder/854954/3/smb/ : SUCCESS in 50m 49s","accounts_in_message":[],"_revision_number":3},{"id":"8b1a97c841592ecd9fe1f432901909271602872e","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-10-11 03:58:28.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/25fad96c3b2443f48fa2c21b9ab8bf7f : SUCCESS in 1h 01m 22s","accounts_in_message":[],"_revision_number":3},{"id":"01038c28449c95a397b50548891c8b1affc57d54","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-11 04:31:33.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/54/854954/3/check/tempest-dsvm-full-focal-py3/efcc9f6/ : SUCCESS in 1h 18m 01s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/54/854954/3/check/devstack-plugin-ceph-dsvm-focal-py3/df6cecf/ : SUCCESS in 1h 32m 24s","accounts_in_message":[],"_revision_number":3},{"id":"0538adf825f0af6fdfeaa717fbdc7dfce925fbbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 04:55:29.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/16dd71980c15407e90b57381410ee8c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b0f68505c1b46f2801053a02a47e89f : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2000ff70c1dd4fa78d6d2a58b20d9502 : SUCCESS in 10m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7bb2c06101cb481e8472a0163e65262f : SUCCESS in 9m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cd9f5a3f29a46999c873aedfffce5d7 : SUCCESS in 10m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/82dab54575f84318abd970f08722341b : SUCCESS in 45m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b9486b2cf3f145d68badc7bb9c5a8f7f : SUCCESS in 56m 29s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7abd33a13e7e47199a70c563e2e89b3c : FAILURE in 24m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b8a7221dac14dc191f1c4d4743cff81 : SUCCESS in 48m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cc52d845088a40aa8b8828c4a2b10acf : SUCCESS in 14m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/eb22c84ba8df4a22878c957e950db87a : SUCCESS in 5m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7c1254d8c2004fd997bb98df0bb51525 : FAILURE in 6m 45s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/10b3f3bf33ca43ce8ca7328e4f119c02 : SUCCESS in 6m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6997999b41094967b7a6cad2175ca501 : SUCCESS in 6m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d01192934e2f46fb86cb0ad5f5d0d0bd : SUCCESS in 32m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2e731e276bd04176b31024508d168d94 : FAILURE in 6m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0d89857df89c454583bf751ad9df8306 : SUCCESS in 1h 04m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/01a5fbdaf00c4849962a590ca64ee292 : FAILURE in 1h 22m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dcd39ad1c3444842a9103f2cce6bf186 : SUCCESS in 1h 41m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/63e00f3fd6fd4d44921dc55faad38ed7 : SUCCESS in 1h 00m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/73349a333dbd47838228ce8dd934f16b : SUCCESS in 1h 12m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/deccc00c5637471bb269b2f1f0350b6d : SUCCESS in 39m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d0952cdf1a5a4cd98efed9f8013e9c46 : SUCCESS in 1h 24m 42s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5fab0c3d0c154b5bb396441237b863ab : SUCCESS in 1h 55m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c4267c385464e6ea1cfed3a30a5a96f : SUCCESS in 53m 33s","accounts_in_message":[],"_revision_number":3},{"id":"862626fa168d44470e217291648f4967c63e9430","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-10-11 04:59:30.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/54/854954/3/check/dsvm-tempest-cisco-zonemanager-job/6277ff7/ : FAILURE in 1h 17m 13s","accounts_in_message":[],"_revision_number":3},{"id":"1bc506bd44143b3e2cfe435e294f6c8e19aec282","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-11 05:08:54.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/54/854954/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/26fb546/ : SUCCESS in 1h 36m 22s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/54/854954/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/2ba6070/ : SUCCESS in 1h 41m 38s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/54/854954/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/dcdef7b/ : FAILURE in 2h 11m 31s","accounts_in_message":[],"_revision_number":3},{"id":"e09000d052dc01faeb59eb93d3cd9cd671ace0ee","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-10-11 05:27:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f26fa83b27ab6274eb7cca380bd040cefde4a6c6","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-10-11 06:01:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d74506e558426064414c6536a43ec1eef33aaf2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 06:27:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/225c8ce23dae4135ba22d2046b13fb56\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/813a8c721da7489eba4bf7434f38d7e8 : SUCCESS in 15m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5a1b714fb5344437893ed5fe0723edff : SUCCESS in 14m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f3d8090841b1fc23a717a4cda9aba913b3a53b50","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-10-11 07:00:59.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/fa1aca49525b404289a875ffcba00f6a : SUCCESS in 58m 41s","accounts_in_message":[],"_revision_number":3},{"id":"fab17e8c3863ec684d340ccb691a5715c42a8663","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-10-11 07:13:26.000000000","message":"Patch Set 3:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/54/854954/3/check/fujitsu-eternus-dx-iscsi/418ad05 : SUCCESS in 1h 45m 31s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/54/854954/3/check/fujitsu-eternus-dx-fc/4eadbc4 : SUCCESS in 1h 41m 34s","accounts_in_message":[],"_revision_number":3},{"id":"d6efc538b4af2bac13ca759046f7c85f7b1ea3e8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-10-11 07:22:55.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/54/854954/3/check/dsvm-tempest-cisco-zonemanager-job/a2231a3/ : FAILURE in 1h 04m 59s","accounts_in_message":[],"_revision_number":3},{"id":"5ebf6fb0d7c10cd391a6902612850e187d0796ab","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-10-11 07:23:50.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/854954/3/iscsi/ : FAILURE in 1h 21m 44s\n- cinder-smb http://cloudbase-ci.com/cinder/854954/3/smb/ : SUCCESS in 57m 21s","accounts_in_message":[],"_revision_number":3},{"id":"a5b1d578f48f1fd36ceb3ae0f8849ce53367cb2c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-11 07:34:26.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/54/854954/3/check/tempest-dsvm-full-focal-py3/89451ec/ : SUCCESS in 1h 24m 08s\n- devstack-plugin-ceph-dsvm-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/cinder/54/854954/3/check/devstack-plugin-ceph-dsvm-focal-py3/9f5e875/ : SUCCESS in 1h 29m 53s","accounts_in_message":[],"_revision_number":3},{"id":"2b6763149080fa8ec61c4e826005c18c6e9e22f1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-11 08:06:29.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/54/854954/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d2d537b/ : SUCCESS in 2h 03m 24s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/54/854954/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/1543a3b/ : SUCCESS in 1h 32m 09s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/54/854954/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/7ed74aa/ : SUCCESS in 1h 29m 28s","accounts_in_message":[],"_revision_number":3},{"id":"2007ca6bbc5ba91b7cee18c09195c16757b2c907","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 08:08:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c6c6c489a944767858d0980d2d3ce2b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd1c0f40bfb64f2e99821f5f625e7426 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7303f5bd42a4cbbb11bbb13e0264883 : SUCCESS in 8m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54bbdcd22a6b42b88de78e4b58af7cda : SUCCESS in 8m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dfbc4e06b9043b3bd6964a7a0577bba : SUCCESS in 11m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/6d8fbbe3c82d476190467846fdbd2bae : SUCCESS in 42m 33s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7a619ee034c04a958966eb2d0d159860 : SUCCESS in 44m 04s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0d69a2706b10479b98ade9d38e968cac : SUCCESS in 1h 02m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9386f684cc594caea9c2e00d64fa1fba : SUCCESS in 42m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/be9b788540014c3289c7b7bbbc68803c : SUCCESS in 14m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/26e2d43554344c80a85468cfc2ef5705 : SUCCESS in 6m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/513313db7a9a4d33beae08fac41f5d4e : FAILURE in 7m 03s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/23e9bdf8e9ef419a8493bd7d226024b8 : SUCCESS in 6m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a53134c24b93480ea0275f8fe4af2f1e : SUCCESS in 6m 53s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e2232422bc3148fe992ef9ba08df475f : SUCCESS in 32m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e7be605ee1d54101a301a08344c44b78 : FAILURE in 6m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/be174517fbac49c78c99e90ae6b31d75 : SUCCESS in 1h 11m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8d83a2903c294201950c01f34d28fb92 : FAILURE in 1h 31m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e174e3f32a3442359b4a90933d7e4351 : SUCCESS in 1h 31m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/05600bb62ffc46f78edd8a04ba3396bb : SUCCESS in 58m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1d68bbb310d946909256f1bff45b8a40 : SUCCESS in 1h 24m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c2eb03c7bf86401abe49ed42dc478579 : SUCCESS in 33m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/825f4c748b1a41748b3b31ae4d58d87e : SUCCESS in 1h 18m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4f513175bb70428e9ddff77fc84b1089 : SUCCESS in 1h 58m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57958c8c7b8448e09ad6d944f01d1447 : SUCCESS in 45m 42s","accounts_in_message":[],"_revision_number":3},{"id":"294c0f295344b545fbf7597df4dd13055f96b2dc","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-10-11 08:26:18.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-54-854954-3/1665475725 : SUCCESS in 17m 30s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":3},{"id":"a64437e369ff7122b02594effe570331646354d7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-10-11 08:36:36.000000000","message":"Patch Set 3:\n\nUNSTABLE build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/749","accounts_in_message":[],"_revision_number":3},{"id":"8bad89063a10efd33224a0987739626bf2fc9ba2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-10-11 09:02:47.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/54/854954/3/check/DellEMC_PowerFlex/997c360/DellEMC_PowerFlex/1410 : SUCCESS in 53m 35s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"8b080e34ad6c10f5840e7b98eb918fcafd0aa390","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-10-11 09:12:13.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/854954/3/upstream-check/cinder-tempest-plugin-ontap-fcp/3f1b62c/ : SUCCESS in 57m 53s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/854954/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/7a8e9e9/ : SUCCESS in 43m 30s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/854954/3/upstream-check/cinder-tempest-plugin-ontap-nfs/6c2de9a/ : SUCCESS in 41m 30s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"b3151ac84f0dfb2f985a2f21f63e152367863526","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-10-11 09:31:07.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/854954/3/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/0ab5cbe/ : SUCCESS in 1h 20m 47s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"c50ba3c3c6ed2e103a201b125d738e8621a60d8f","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-10-11 09:31:51.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/54/854954/3/check/DellEMC_SC_FC/d8ed12c/DellEMC_SC_FC/1056 : SUCCESS in 55m 50s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/54/854954/3/check/DellEMC_SC_iSCSI/5094109/DellEMC_SC_iSCSI/1040 : SUCCESS in 1h 22m 40s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"6321984cb574e258e8ea52159d1d98e4b067e795","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-10-11 09:53:02.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/54/854954/3/check/DellEMC_PowerStore_iSCSI/fe4e75f/DellEMC_PowerStore_iSCSI/840 : FAILURE in 1h 24m 15s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/54/854954/3/check/DellEMC_PowerStore_FC/d634c34/DellEMC_PowerStore_FC/992 : SUCCESS in 1h 04m 05s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/54/854954/3/check/DellEMC_PowerStore_NFS/5319580/DellEMC_PowerStore_NFS/531 : SUCCESS in 41m 02s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/54/854954/3/check/DellEMC_PowerStore_NVMeTCP/7aa832e/DellEMC_PowerStore_NVMeTCP/402 : SUCCESS in 1h 43m 46s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"27ca6ef04d662e006aaa797d5c5499f68552a0de","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-11 10:05:05.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/54/854954/3/check/DellEMC_Unity_FC/358b341/DellEMC_Unity_FC/1921 : SUCCESS in 1h 12m 51s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/54/854954/3/check/DellEMC_Unity_iSCSI/112359f/DellEMC_Unity_iSCSI/1768 : SUCCESS in 1h 55m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"05f9db2ec085d9056c61c2e31bba937bce82ca28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-11-21 01:13:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fb96863e798158bfe3b9e9650040c863d28ff2fe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-21 01:13:37.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1756 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"ccb2c3131dc62adbd3b4d45927eaee6cc60f82da","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-21 01:19:39.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"232925d84af28a0a9cdd87bc98235a30f9802a94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 01:35:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed09dad5a2d94039a6b546581bbfd458\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/999f51c71c45439faf19a4040befede2 : FAILURE in 15m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d75e75b41008487c8cd9a719720bc032 : FAILURE in 15m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ba6f5d8729b2c2c1f4cc2a55b772a8a1b07435c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 03:40:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/efd995b20c44490e8105462465ce9b23\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a244d997c48049d8851939df4016ec2b : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a178349d596d495aa82b50f31157fde2 : FAILURE in 10m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5320cdd4616441f882bcc72936bb18f6 : FAILURE in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/000cd2eb16614114bd170add59937d6f : SUCCESS in 13m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/a4b98ae9c15d43f1b76936f74cbd745a : SUCCESS in 1h 08m 03s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/06a57118dac84d46bfe14fe045ccc023 : SUCCESS in 52m 09s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/34702d67d42b422097bc556c8d47f7f6 : SUCCESS in 1h 27m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63eed67002784a1fa963e40756d67e34 : SUCCESS in 58m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9fd820100e6f40c7a0dd83e93a1bc063 : FAILURE in 13m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/41c0fcb502894f5a8197e6e22d3469e4 : SUCCESS in 5m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/41df35ab97bc4abda471ac21c27ab61b : FAILURE in 6m 51s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0b17719844444e84ac91bfd61ca1351f : SUCCESS in 6m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/49ac2d90378e40c1a47828a7d858526d : SUCCESS in 6m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b659ab6bcd9e402dac6fb046d10bf4bd : SUCCESS in 37m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b68949f50ca4438891b0dbccfb716929 : FAILURE in 5m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/899bf38946794aabb42521f65530824b : SUCCESS in 1h 23m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b6fd58a58167437b967d83afb0e2f82b : FAILURE in 1h 22m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b0b9802b2f624538ab6455b9c04abeaf : SUCCESS in 2h 19m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fef0cbc763d8406f89fe6ff7bae4c6d1 : SUCCESS in 1h 08m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/14a2f1c8b18a4874bd4897370e802dd4 : SUCCESS in 1h 22m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a8c23a69162a45c3a507b0edf2e3213a : SUCCESS in 32m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c2828ff234a945eaad0320ddd365ec41 : SUCCESS in 1h 26m 52s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/29b825bc3d6c4df38f71e9382db8a487 : SUCCESS in 1h 55m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d0568ee352e41ddbf84860907cd1c34 : SUCCESS in 40m 07s","accounts_in_message":[],"_revision_number":4},{"id":"38b7ea5a29474f2e1058feb44775b3126953b03e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-21 04:32:54.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/54/854954/4/check/iscsi-driver-alletra-6k/87ddc48/87ddc48.html : FAILURE in 1h 39m 42s","accounts_in_message":[],"_revision_number":4},{"id":"767e6a65787429815297bf56e3491e442647e47d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-21 06:05:26.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-54-854954-4/ns5_nfs : SUCCESS in 3h 28m 25s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-54-854954-4/ns5_iscsi : SUCCESS in 3h 37m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"69d048903cd7aa9332f16f8c82bc91f87b870b31","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-21 06:06:57.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/854954 : SUCCESS in 1h 50m 34s","accounts_in_message":[],"_revision_number":4},{"id":"c14be1183c177d6c4198742c3205f3c1e9b4437c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-21 06:21:38.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/54/854954/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/9aac501/ : SUCCESS in 1h 42m 08s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/54/854954/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3d94658/ : SUCCESS in 1h 57m 33s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/54/854954/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/92fe67a/ : SUCCESS in 1h 23m 57s","accounts_in_message":[],"_revision_number":4},{"id":"8051c170702e81cd340dc8da33de03ea4fa29f8c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-21 06:22:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/854954 : SUCCESS in 2h 00m 28s","accounts_in_message":[],"_revision_number":4},{"id":"5524745778c7aa9b74f55fc8f78915bc4ebeb303","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-11-21 06:45:58.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/54/854954/4/check/fujitsu-eternus-dx-iscsi/18a4f1a : FAILURE in 1h 45m 32s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/54/854954/4/check/fujitsu-eternus-dx-fc/d3a0d70 : SUCCESS in 1h 49m 30s","accounts_in_message":[],"_revision_number":4},{"id":"3e81e285b980ded65a5584300bdefbf22fcce3cd","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-11-21 11:25:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/54/854954/4/ : SUCCESS in 1h 33m 02s","accounts_in_message":[],"_revision_number":4},{"id":"2134263226a4178befda7db1a0f357891a0851ce","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-11-22 04:31:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/854954/4/20221122064353/NEC-FC : SUCCESS in 3h 32m 08s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/854954/4/20221122064353/NEC-ISCSI : SUCCESS in 6h 47m 47s","accounts_in_message":[],"_revision_number":4},{"id":"618a132cfb64607f1ba6b44389f1922f2473d3d8","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-11-22 06:37:54.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/854954/4/20221122093042/NEC-V-ISCSI : SUCCESS in 2h 32m 06s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/854954/4/20221122093042/NEC-V-FC : FAILURE in 6h 07m 08s","accounts_in_message":[],"_revision_number":4},{"id":"68f17a1a8a8226940270caf3d565a7e6e881c458","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-11-22 08:12:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3f5b88106d7842620dc98c6139a404bd500dc650","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-22 08:12:50.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1766 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"8f1bf8bb912fd353bbcd07744ad7735570f70992","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-22 08:14:24.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"c38db4bd31feb06df25145bd7b04293a2e5bc199","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 08:36:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c2eab3a5de24e2b9c432e6f79690bd0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c3994800992470497d933c2f2dedd0a : SUCCESS in 22m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5c0a6e3ba83a457fa79d3a1edb5b0b1a : SUCCESS in 21m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"06137c11531b04443a435e02929d7d5d551ede98","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-22 08:40:01.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/54/854954/5/check/iscsi-driver-alletra-6k/10d8dd1/10d8dd1.html : FAILURE in 25m 38s","accounts_in_message":[],"_revision_number":5},{"id":"1554c7174317651c00b7fe04c4e25f37131d8a97","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-22 09:19:31.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.157/logs/54/854954/5/check/dsvm-tempest-cisco-zonemanager-job/6c2f183/ : POST_FAILURE in 12m 02s","accounts_in_message":[],"_revision_number":5},{"id":"46ded13df60cb2a30008757a4d435064b4473b36","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-22 09:41:21.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/854954 : SUCCESS in 1h 26m 40s","accounts_in_message":[],"_revision_number":5},{"id":"9ce951c9a65f675278b3f1cc7578d2f6c5926b96","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-22 09:55:34.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/854954 : FAILURE in 1h 41m 32s","accounts_in_message":[],"_revision_number":5},{"id":"eba9e870dd0d033667d9e8be3d07f0d99564f0ed","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-22 10:04:58.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/54/854954/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f9877b5/ : SUCCESS in 1h 25m 13s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/54/854954/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/e7ef492/ : SUCCESS in 1h 51m 26s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/54/854954/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/00a9e86/ : SUCCESS in 1h 38m 52s","accounts_in_message":[],"_revision_number":5},{"id":"98f3b9c9318e095067b8379e6f7323f5bfe2ce05","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-11-22 10:15:27.000000000","message":"Patch Set 5:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/54/854954/5/check/fujitsu-eternus-dx-iscsi/b28046f : SUCCESS in 2h 02m 07s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/54/854954/5/check/fujitsu-eternus-dx-fc/fe010ff : SUCCESS in 2h 00m 04s","accounts_in_message":[],"_revision_number":5},{"id":"3ea3f9820f2949c621410d06135cb61c192e9927","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 10:19:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5be2af1fbdcd433fad2a941c0873ca4d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/484e8edd36b44f06aa861442fb6827d0 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8291f521484a48afbf77b5500b10dda2 : SUCCESS in 11m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec91b00cacec4ef9ac0b0b9618603e52 : SUCCESS in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0f00697671a4644b586bd1b91349812 : SUCCESS in 10m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/8b9f9dc55a554e9bb9490ecab860486f : SUCCESS in 42m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/dfa65fcd12f948a48abd3865dbcde294 : SUCCESS in 49m 20s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/07856e002daf4e889dccf9a00ad3481f : SUCCESS in 1h 18m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5fe1b3a9677c4fffa91c3073c43dada5 : SUCCESS in 42m 43s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7915acd56d984fc68d431c20f1267d56 : SUCCESS in 13m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6f98e65ee7954cd08b0d39e13d0c8a13 : SUCCESS in 4m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fe00391a86ac47b7a7430921d99e9fa6 : FAILURE in 6m 14s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b0736bf985414e66899ab5889a06fb0c : SUCCESS in 6m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4955ad0fa1c34eefb9da55d06d399db4 : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6062ed1425c2422284ccb9418f6eab38 : SUCCESS in 32m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a447bfc9c75d4093a47153b236cf749e : FAILURE in 5m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/49b8f766f2014f0badd57347d897eb1f : SUCCESS in 1h 01m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2fa3c889e1e04f5db015322fde394209 : FAILURE in 1h 09m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/22d8cc850b06418e8a21c95b766c57ef : SUCCESS in 1h 37m 22s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/42e346b9595d49c1abadfd786304cb35 : FAILURE in 52m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e74ce8d33c984e5cb3516ade8aedb89f : SUCCESS in 1h 37m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8ea60ab8d2a54a169574ec9eacee74fc : SUCCESS in 40m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a177a1a822614be8a239a191afd5b2e9 : SUCCESS in 1h 17m 45s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f00690c4f9564ec08db0391da71583e7 : FAILURE in 2h 02m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18f7c0bd634b49ce8e36d10a4d6d3a36 : SUCCESS in 35m 12s","accounts_in_message":[],"_revision_number":5},{"id":"89dce9487551369164ccfac6265bf475020929cc","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-22 11:45:26.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-54-854954-5/ns5_nfs : SUCCESS in 3h 27m 16s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-54-854954-5/ns5_iscsi : SUCCESS in 3h 32m 13s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"e594e849763d507814060fdcdc4ed00861aed2bb","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-11-22 12:39:54.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/54/854954/5/ : SUCCESS in 1h 37m 26s","accounts_in_message":[],"_revision_number":5},{"id":"00a922d829548d253dcc1180d9eb5f4a5fcc5dec","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-22 16:35:25.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/54/854954/5/check/DellEMC_Unity_FC/45fc692/DellEMC_Unity_FC/2022 : SUCCESS in 56m 10s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/54/854954/5/check/DellEMC_Unity_iSCSI/edef430/DellEMC_Unity_iSCSI/1869 : FAILURE in 2h 08m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"ffeb8c4000f251fcb3a5508216088c232349833f","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2022-11-23 01:26:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0fa6dc02e4e21690a708b10fd6c3166132bce06f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 01:44:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7b16a29e17d4228a7be4a68fb7ab0ae\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9448accee6c043d883e8c4860a8fd026 : SUCCESS in 13m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5a22bd1d4521403bb2ef840ba045d505 : SUCCESS in 13m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"14dbf71cf00af0ad26623aac0adee04d819866f9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-23 01:48:23.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/54/854954/5/check/dsvm-tempest-cisco-zonemanager-job/cb5565a/ : POST_FAILURE in 12m 38s","accounts_in_message":[],"_revision_number":5},{"id":"caeab2f62ab8056913a21ec60337cb8899f0441a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 03:26:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dcc09af0daa445782466b8bb5ae4c62\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff6ca32543b24f9e8ed1327e90cdafdc : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01b33c74f31d4e30a11cda7c1aa38544 : SUCCESS in 10m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/859c39ee07174300843f1bde4d0a798d : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dc16b9b097443cc8929777cf1a7d4a5 : SUCCESS in 12m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/3799248421d144dfa30a0af6f6af3609 : SUCCESS in 51m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a9f1f6b21a2145f3a11b838e051d7df9 : SUCCESS in 52m 44s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/23b1f816d1ce40dca1cda1e32adfaed6 : SUCCESS in 1h 31m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22b21075524840c99dd0c271ed821138 : SUCCESS in 49m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/92bc9113116a4e77aff4a7a155367638 : SUCCESS in 13m 40s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b4c5f72ae6cb4e829676724ed0325f4a : SUCCESS in 5m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/27ab77a30fa24261b049d0a5eff385f1 : FAILURE in 6m 18s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/71006ccff9574b129f7c4a0b926ada65 : SUCCESS in 6m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9844f05f4cd545148dfa4eb73f00048c : SUCCESS in 6m 05s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2539e8d035a542ea92d8bc5a72491ff8 : SUCCESS in 1h 06m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b374cf11dd3a46d3a4f7409b39dca32a : FAILURE in 5m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/29cff89233954370b6926feb48d1cf19 : SUCCESS in 1h 20m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/91fb0eeb2f3749bfbaf4aca927279573 : FAILURE in 1h 25m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/56139c5a65fd42e5b589f3579d9183bc : SUCCESS in 1h 52m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/85c1ce0aef774a64b77f3f5dcf4a94e5 : SUCCESS in 1h 07m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/37c3ebe0422640baba9a6607cdf1a55a : SUCCESS in 1h 19m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bdd8400d802f423caa8ee9e818ec5b86 : SUCCESS in 33m 31s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/205c280a6227436592e5d582bf063206 : SUCCESS in 1h 20m 41s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9ce30594041f4192aee699cef18596b4 : SUCCESS in 1h 57m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44fa424de7694e199eceb227f203d730 : SUCCESS in 46m 52s","accounts_in_message":[],"_revision_number":5},{"id":"e3fc7650acc73d5a5d68955f891a0bfe1a4b992b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-23 03:33:53.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/54/854954/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/efb2113/ : SUCCESS in 1h 36m 34s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/54/854954/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6b8a7fb/ : SUCCESS in 2h 05m 54s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/54/854954/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/2c72bd8/ : SUCCESS in 1h 37m 49s","accounts_in_message":[],"_revision_number":5},{"id":"5739864c0007aa9a75be697d2b8351ea56446470","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-11-23 04:20:04.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/854954/5/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/2fc1d87/ : SUCCESS in 52m 06s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":5},{"id":"eb482fe79bbe34a9e91024292c4e8269df742c35","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-11-23 04:29:24.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/854954/5/upstream-check/cinder-tempest-plugin-ontap-fcp/c3fa8ea/ : SUCCESS in 57m 33s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/854954/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/f0a4826/ : SUCCESS in 42m 24s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/854954/5/upstream-check/cinder-tempest-plugin-ontap-nfs/1c0a5d5/ : SUCCESS in 44m 07s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"6c2cebecb15199cf9795b1d6504f5053a3476cfd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-23 05:13:50.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/54/854954/5/check/DellEMC_Unity_FC/cee83a9/DellEMC_Unity_FC/2025 : SUCCESS in 1h 06m 25s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/54/854954/5/check/DellEMC_Unity_iSCSI/42ad452/DellEMC_Unity_iSCSI/1872 : FAILURE in 1h 47m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"f5d56f2feef14babdb22e1109b35e61ffb8cf8e3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-11-24 01:00:33.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/854954/5/20221124031634/NEC-FC : SUCCESS in 3h 31m 59s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/854954/5/20221124031634/NEC-ISCSI : SUCCESS in 6h 43m 56s","accounts_in_message":[],"_revision_number":5},{"id":"b4f221c6f9850069f12b2d882a651781f9427b06","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-11-24 08:09:35.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/854954/5/20221124112709/NEC-V-ISCSI : SUCCESS in 2h 35m 55s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/854954/5/20221124112709/NEC-V-FC : SUCCESS in 5h 42m 23s","accounts_in_message":[],"_revision_number":5},{"id":"2d2c29c5fb147645e5c5026d99d02570c8d0b944","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-03-22 03:25:51.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"74961d001c1313f362480b430864243b9e62dab1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-04-06 07:24:41.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"77743c9225a00bfbca63af4ff17ebe43b0b36837","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 07:38:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4757a4377a044057a7520ac7eebacc00\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f733ff0d6fcb4c6dbe7ffd9e4c658d64 : SUCCESS in 12m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5e0c0e5006fb4bb4b1679572418601cb : SUCCESS in 8m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2b7ac7cc87552fcabe3393c92ec7fb5105893cdb","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-04-06 08:21:46.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/77a2d068010d4a10acc5ba5e71db81ae : FAILURE in 56m 13s","accounts_in_message":[],"_revision_number":6},{"id":"20e9a2eb676101a0541f67ad63bbd96d3c26281b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-04-06 08:22:37.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/bc4021101dc74b89b34fb03fca98b184 : SUCCESS in 56m 14s","accounts_in_message":[],"_revision_number":6},{"id":"d5881263fe5d11716aa329a3d0bd48648c538b8e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-04-06 08:30:48.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/54/854954/6/check/dsvm-tempest-cisco-zonemanager-job/f053e69/ : FAILURE in 1h 03m 48s","accounts_in_message":[],"_revision_number":6},{"id":"b29d951756c3144c5612ea2ac040a14789e2a475","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-04-06 08:50:19.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/54/854954/6/ : POST_FAILURE in 1h 17m 36s","accounts_in_message":[],"_revision_number":6},{"id":"320ba720bca4b65e6dbfa2b7289ed55d211f7b09","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-04-06 08:59:41.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/54/854954/6/check/fujitsu-eternus-dx-iscsi/c8458c9 : SUCCESS in 1h 34m 19s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/54/854954/6/check/fujitsu-eternus-dx-fc/78598af : FAILURE in 26m 37s","accounts_in_message":[],"_revision_number":6},{"id":"3330391c3462d10652dbee669f8f5f7143e43f5c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-04-06 09:00:40.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/854954 : SUCCESS in 1h 34m 14s","accounts_in_message":[],"_revision_number":6},{"id":"d097e6fea812fff63c075586482b796a719c64fd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-06 09:26:00.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/54/854954/6/check/iscsi-driver-alletra-6k/1c61895/1c61895.html : FAILURE in 2h 00m 07s","accounts_in_message":[],"_revision_number":6},{"id":"e27c6746fa0142f49af81bb6f511d48b48764d62","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-04-06 09:26:36.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/54/854954/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/0e744c8/ : SUCCESS in 1h 46m 03s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/54/854954/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/473c980/ : FAILURE in 1h 59m 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/54/854954/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/6b634f6/ : SUCCESS in 1h 44m 08s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/54/854954/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/467e411/ : FAILURE in 34m 50s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f314533d443c1ff15797dfdfb23d4a2f6aeed564","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-04-06 09:27:40.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d4b1b70f40a6e692e0a6beebfa6ed700371337c2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-04-06 10:14:19.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-854954-6/FC/ : SUCCESS in 2h 11m 28s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-854954-6/iSCSI/ : SUCCESS in 2h 49m 17s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"e5c136233f46d1a42af974a6d06a71bc74b0b82f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-04-06 10:40:04.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/854954/6/20230406164836/NEC-V-FC : SUCCESS in 2h 42m 50s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/854954/6/20230406164836/NEC-V-ISCSI : SUCCESS in 2h 51m 25s","accounts_in_message":[],"_revision_number":6},{"id":"c22439988c3a634e4e87d8fbff296cbab26d1b89","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-04-06 11:02:17.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-54-854954-6/ns5_nfs : SUCCESS in 3h 37m 4s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-54-854954-6/ns5_iscsi : SUCCESS in 3h 25m 57s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"15e8b3ed726f1b5d666198b18be16209da4a385b","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-04-06 11:09:09.000000000","message":"Patch Set 6:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/54/854954/6/check/fujitsu-eternus-dx-iscsi/7a1e355 : SUCCESS in 1h 40m 55s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/54/854954/6/check/fujitsu-eternus-dx-fc/83641e3 : SUCCESS in 1h 34m 42s","accounts_in_message":[],"_revision_number":6},{"id":"492143937c09d3c510c84fb4929b677d0e7bf581","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 11:09:41.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad36ca522e35494d9060e1613bf40c3a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bb593d638ef41eba8d75434403797fc : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2b414d4916448d9ae566571c6c68a4c : SUCCESS in 15m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61a7aaf47b17403d9f54565e1955387f : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbda6ec52cea44aca9578f9d5ce7c0e7 : SUCCESS in 11m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/63c89f9111c74aa0b6efcc3852a7290e : SUCCESS in 49m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/91e61fb0805e45e194c2ff86180dde40 : SUCCESS in 1h 18m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f717a85c39e64422aacffd8800e83b3c : SUCCESS in 51m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b64446ae5d143cdb595520417658a5e : SUCCESS in 9m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/154f5b7b099e4bcd99f82ec2dceba169 : SUCCESS in 16m 40s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b559a294953943a9aedc24cca72c54ce : SUCCESS in 4m 26s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c1ab7b45bf704feb9d7e4e28b97b800c : FAILURE in 6m 21s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ffee6f4b35d04744b8776cda4b8fe711 : SUCCESS in 5m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/165d37a9581043c7b4f244b7fd1d2f20 : SUCCESS in 4m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a55b736658c94a03b78826454e0fc6ba : SUCCESS in 31m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7efd4083b3644dfc9b5bc15d024d9a4f : FAILURE in 5m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9827d721e0e0436fb3123dc9f0b1af02 : FAILURE in 1h 10m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4ef04773df874daf9238d3bcf283fc84 : FAILURE in 1h 22m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9bb464dee1834627bff141245ced6f6c : SUCCESS in 1h 27m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ef4f361f6595450390fb42bb1f21fccc : SUCCESS in 54m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/da8f6bcdd0d942c09532206dd015b433 : SUCCESS in 1h 16m 49s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f9a31cf0c3cb452a8dcd42278fd76e70 : SUCCESS in 43m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9deecdd170ec49c4b658302a83fff6ea : SUCCESS in 1h 01m 23s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e801e12596784ccf982555cdbf0e9724 : SUCCESS in 2h 02m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ccdaf9f83ee041eeb99a0d1909cddf56 : SUCCESS in 39m 03s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6b2ed125c844f43948dbe18ccc6314c : SUCCESS in 1h 05m 51s","accounts_in_message":[],"_revision_number":6},{"id":"82ab4010b544b6ac754defa5308938e84d8029ec","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-04-06 12:03:18.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/54/854954/6/20230406172845/NEC-ISCSI : SUCCESS in 3h 21m 56s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/54/854954/6/20230406172845/NEC-FC : SUCCESS in 3h 34m 30s","accounts_in_message":[],"_revision_number":6},{"id":"208af5c8b411d5a8b3cb083d22411a03c2607afe","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-04-06 12:13:14.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/54/854954/6/check/tempest-dsvm-full-focal-py3/472d824/ : SUCCESS in 1h 31m 13s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/5ed976d61e9f4f53832748ebb45fbf87 : RETRY_LIMIT in 1m 04s","accounts_in_message":[],"_revision_number":6},{"id":"23471309034711efb5f313f698d5233a39943c5e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-19 00:20:21.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/54/854954/6/check/iscsi-driver-nimble/59d90fc/59d90fc.html : FAILURE in 1h 44m 46s","accounts_in_message":[],"_revision_number":6},{"id":"6b04b5eed2c269021abbc66c13cb5871fedb1c93","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-05-23 02:27:57.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b21dc187ca2d45320bbe9f5d3c1d73f5b786d353","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-05-23 02:29:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a951f093d25c30eb2678175fd2cc15f6e593af5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 02:59:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bae7a1da8e6b4ce888cf79ae6578cd79\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49841f92e1ed4dbaa816800ad67d696d : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/63e54f56dcc541e0969d553d6c9c42cc : FAILURE in 27m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/774ee65f479847ac845246e8bf26b1c2 : SUCCESS in 10m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"045254b6cfbf8b16feffc39e22f5ed0791f2114d","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-23 03:18:31.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/5cdbeedd1b90460fbf457ba78f21009d : POST_FAILURE in 49m 45s","accounts_in_message":[],"_revision_number":6},{"id":"32ec238d1e47f6538f0351c9e89ada7b79c2f986","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-23 03:28:59.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0e3fb86fe8fc41bd8334723041f4ff6c : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":6},{"id":"d6e2fefc77fc28bcde0d468e1bb7827b4b26757c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-23 03:41:52.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/fb73733d5fee4f70ab1dbe0aba15321c\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/c473c793dc534eaf9c015770bd8d402e : FAILURE in 1h 10m 40s","accounts_in_message":[],"_revision_number":6},{"id":"02242327e643f8e307fb38fdb2b7fdf27e12a4d7","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-05-23 04:16:10.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/54/854954/6/check/fujitsu-eternus-dx-iscsi/a3779ec : FAILURE in 1h 46m 25s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/54/854954/6/check/fujitsu-eternus-dx-fc/241b5a9 : SUCCESS in 1h 37m 07s","accounts_in_message":[],"_revision_number":6},{"id":"e3856a311098e7337b49da3dda98603e20f674a5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-05-23 05:00:51.000000000","message":"Patch Set 6:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1363/","accounts_in_message":[],"_revision_number":6},{"id":"872ab381af3341763a5444fcaf4e9d2986bde541","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-05-23 05:04:55.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ec25e536957baec14504fed27aabcf23467d5633","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 05:16:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66936a7bd5774803af618354934c9a5d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61b7061406df44ce845ed5f8260dc5d3 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/630a3cff8f184a5f9ef81bfe25ea831e : SUCCESS in 11m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97e974bf86dc435fb9d48c6a2bef955f : SUCCESS in 14m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8374816906ed41fa97357231fafb8dae : SUCCESS in 12m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d9705392e464412aa104a1683563ac1 : SUCCESS in 10m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/97de5de887c4400bbbd9d98c8b438c44 : SUCCESS in 50m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c01d773ec8734305b5bee92e31fce776 : SUCCESS in 1h 18m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2d535bbc10e42be9019a1d597c56a6d : SUCCESS in 50m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f40fd9415664d829601e8aac2ce0f5f : SUCCESS in 7m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1e639278625d4bce9736b9f1e30c1cf8 : SUCCESS in 16m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3fbac66795ff49079cd8c325f617feae : SUCCESS in 7m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c422125de4734c27b6a1a8bbe044bd76 : FAILURE in 6m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6b0616d53b894b8288d54403d217fd0b : SUCCESS in 7m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c0d82b5817aa43edacbe73bba0864df7 : SUCCESS in 6m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f824fc25621248e58c652353bf1a9338 : SUCCESS in 41m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e35da69e2fcb45f89ba04223197e47f6 : SUCCESS in 7m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/30245e9a35134007afb6448d99aa057d : SUCCESS in 1h 44m 35s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c5aebd4a03c44209b31386f195029586 : FAILURE in 1h 37m 26s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f2090835241944b9b33c6e8c82f24143 : SUCCESS in 2h 05m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ea07d9b5bd384d7da6e145dc76956b0b : POST_FAILURE in 9m 55s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1f19a36c527c4c21b5bd9bb8cb0a0427 : SUCCESS in 1h 15m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/607cf7307c544861aac361352b60bc9d : SUCCESS in 1h 34m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dc2f7e65d65044d4a690cc05e53b77b2 : SUCCESS in 46m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/99379a76a3884015ad28a3a3f6cefbfe : SUCCESS in 1h 23m 23s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7ced6a1ce514494b8d5f632bd86dd903 : FAILURE in 25m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/27ef397e1915411fb0878a6d013dd226 : SUCCESS in 2h 29m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72dcdac009494ff1883bb071aa3b5c31 : SUCCESS in 48m 00s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92ac59dd85e94c8d818dc685775d279c : SUCCESS in 1h 32m 00s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/fa0f9fc01ca54311a702ee094ce0c3d3 : SUCCESS in 11m 32s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b3e21e4c79c953074e150ee5cf7fef096b204492","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-05-23 05:18:05.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"3f273ee893266825ae5cc49e4296047170ac9380","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-05-23 05:23:19.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_05_2023_05_16_29\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"c5335b0feb3f99137143aae5cf10d5fb7bbc8244","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-05-23 05:41:14.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"c884e43c1affc88fb3a9bef2bf6669b28365da44","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-05-23 05:41:24.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"855372d925c35e4bb512563e30883073af250e25","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-05-23 06:35:04.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/54/854954/6/check/DellEMC_PowerStore_iSCSI/cfa1f14/DellEMC_PowerStore_iSCSI/53 : SUCCESS in 49m 41s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/54/854954/6/check/DellEMC_PowerStore_FC/84d8ca1/DellEMC_PowerStore_FC/53 : SUCCESS in 1h 18m 21s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/54/854954/6/check/DellEMC_PowerStore_NFS/56847a9/DellEMC_PowerStore_NFS/52 : SUCCESS in 53m 26s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/54/854954/6/check/DellEMC_PowerStore_NVMeTCP/98fe9a1/DellEMC_PowerStore_NVMeTCP/52 : SUCCESS in 1h 03m 19s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"99b7b664dbfc1fd8bfb00e0aedd8c0441737fa3d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-05-23 06:52:23.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/54/854954/6/check/DellEMC_PowerFlex/ac8453a/DellEMC_PowerFlex/97 : SUCCESS in 1h 35m 42s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"104035830f5385a4ab13d6a9316b486f8bd876ff","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-05-23 06:53:19.000000000","message":"Patch Set 6:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/54/854954/6/check/fujitsu-eternus-dx-iscsi/c63205f : SUCCESS in 1h 34m 02s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/54/854954/6/check/fujitsu-eternus-dx-fc/dc87648 : ABORTED in 1h 47m 47s","accounts_in_message":[],"_revision_number":6},{"id":"bc4aea8ebf2fe5390472608a55d4b1a4593a91a5","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-05-23 06:55:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"13e862f62651e6630484f4f51e7f399c19ef2d43","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-05-23 06:56:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0bba29fb9c070d80018555b1a58b1c27172cbde7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-23 07:09:06.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/854954/6/upstream-check/cinder-tempest-plugin-ontap-fcp/1f2c9f3/ : RETRY_LIMIT in 8m 45s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/854954/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/1525533/ : SUCCESS in 1h 46m 08s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/854954/6/upstream-check/cinder-tempest-plugin-ontap-nfs/357dcb4/ : SUCCESS in 1h 46m 43s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/854954/6/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/4ee44c6/ : FAILURE in 1h 43m 56s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"1a7568885fe75c085aff78fed6fd504976ff15fb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-23 07:39:12.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/54/854954/6/check/DellEMC_Unity_FC/0b5c385/DellEMC_Unity_FC/168 : FAILURE in 2h 22m 27s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/54/854954/6/check/DellEMC_Unity_iSCSI/6e0f85a/DellEMC_Unity_iSCSI/169 : SUCCESS in 2h 15m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"ce3252d6905dcf4a29f35691c51a8ed82f75935e","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-05-23 09:08:16.000000000","message":"Patch Set 6:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/54/854954/6/check/fujitsu-eternus-dx-iscsi/e8cd934 : SUCCESS in 2h 10m 53s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/54/854954/6/check/fujitsu-eternus-dx-fc/e6d9ae3 : SUCCESS in 1h 50m 59s","accounts_in_message":[],"_revision_number":6},{"id":"ed3c32741c20d543f1abe7d88093b927bfde6fc0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-09-15 01:16:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6},{"id":"77451a749957f2da3f4ab1c3f4fac567bb529b6b","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-09-15 04:00:07.000000000","message":"Patch Set 6:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/54/854954/6/fc : SUCCESS in 0h 45m 11s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/54/854954/6/iscsi : SUCCESS in 0h 45m 56s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"9ff369bb5b39af90647366552dc165e477f606f1","revisions":{"15ee7184baa4186404f1be83e688595551e0e1c4":{"kind":"REWORK","_number":1,"created":"2022-08-29 08:04:33.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/54/854954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/854954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/854954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/854954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/854954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/854954/1"}}},"commit":{"parents":[{"commit":"2e6c0dfcecf18b107010d6ece21a85f1be4007bb","subject":"Merge \"Bump mypy version to 0.960\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2e6c0dfcecf18b107010d6ece21a85f1be4007bb"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2022-08-29 08:07:51.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2022-08-29 08:07:51.000000000","tz":-240},"subject":"Fujitsu Driver: Add Timeout function","message":"Fujitsu Driver: Add Timeout function\n\nWhen executing some functions, added the timeout for them.\n\nfollowing constants added\n* fujitsu_set_func_timeout\n* fujitsu_query_func_timeout\n* fujitsu_eternus_req_timeout\n* fujitsu_timeout_retry_times\n\nChange-Id: Ia36b8892e88453716f68956c44daeb14d8854912\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/15ee7184baa4186404f1be83e688595551e0e1c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/15ee7184baa4186404f1be83e688595551e0e1c4"}]},"branch":"refs/heads/master"},"d7fa0fe4c0c42127387ec489a644d393139cd1a8":{"kind":"REWORK","_number":2,"created":"2022-08-29 08:55:03.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/54/854954/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/854954/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/854954/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/854954/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/854954/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/854954/2"}}},"commit":{"parents":[{"commit":"2e6c0dfcecf18b107010d6ece21a85f1be4007bb","subject":"Merge \"Bump mypy version to 0.960\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2e6c0dfcecf18b107010d6ece21a85f1be4007bb"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2022-08-29 08:07:51.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2022-08-29 08:58:52.000000000","tz":-240},"subject":"Fujitsu Driver: Add Timeout function","message":"Fujitsu Driver: Add Timeout function\n\nWhen executing some functions, added the timeout for them.\n\nfollowing constants added\n* fujitsu_set_func_timeout\n* fujitsu_query_func_timeout\n* fujitsu_eternus_req_timeout\n* fujitsu_timeout_retry_times\n\nChange-Id: Ia36b8892e88453716f68956c44daeb14d8854912\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7fa0fe4c0c42127387ec489a644d393139cd1a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7fa0fe4c0c42127387ec489a644d393139cd1a8"}]},"branch":"refs/heads/master"},"f599c3d6da6fb3382578e9887b7141ffe715649f":{"kind":"REWORK","_number":3,"created":"2022-08-29 08:58:20.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/54/854954/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/854954/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/854954/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/854954/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/854954/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/854954/3"}}},"commit":{"parents":[{"commit":"2e6c0dfcecf18b107010d6ece21a85f1be4007bb","subject":"Merge \"Bump mypy version to 0.960\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2e6c0dfcecf18b107010d6ece21a85f1be4007bb"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2022-08-29 08:07:51.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2022-08-29 09:02:15.000000000","tz":-240},"subject":"Fujitsu Driver: Add Timeout function","message":"Fujitsu Driver: Add Timeout function\n\nWhen executing some functions, added the timeout for them.\n\nfollowing constants added\n* fujitsu_set_func_timeout\n* fujitsu_query_func_timeout\n* fujitsu_eternus_req_timeout\n* fujitsu_timeout_retry_times\n\nChange-Id: Ia36b8892e88453716f68956c44daeb14d8854912\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f599c3d6da6fb3382578e9887b7141ffe715649f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f599c3d6da6fb3382578e9887b7141ffe715649f"}]},"branch":"refs/heads/master"},"52015d6c818be611069b170cbe93e67ca336d1cc":{"kind":"REWORK","_number":4,"created":"2022-11-21 01:13:28.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/54/854954/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/854954/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/854954/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/854954/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/854954/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/854954/4"}}},"commit":{"parents":[{"commit":"2e6c0dfcecf18b107010d6ece21a85f1be4007bb","subject":"Merge \"Bump mypy version to 0.960\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2e6c0dfcecf18b107010d6ece21a85f1be4007bb"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2022-08-29 08:07:51.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2022-11-21 01:18:03.000000000","tz":-300},"subject":"Fujitsu Driver: Add Timeout function","message":"Fujitsu Driver: Add Timeout function\n\nWhen executing some functions, added the timeout for them.\n\nfollowing constants added\n* fujitsu_set_func_timeout\n* fujitsu_query_func_timeout\n* fujitsu_eternus_req_timeout\n* fujitsu_timeout_retry_times\n\nChange-Id: Ia36b8892e88453716f68956c44daeb14d8854912\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/52015d6c818be611069b170cbe93e67ca336d1cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/52015d6c818be611069b170cbe93e67ca336d1cc"}]},"branch":"refs/heads/master"},"8faaebef1850c7e3f3bc0ee483bb859213eb9f29":{"kind":"REWORK","_number":5,"created":"2022-11-22 08:12:41.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/54/854954/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/854954/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/854954/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/854954/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/854954/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/854954/5"}}},"commit":{"parents":[{"commit":"2e6c0dfcecf18b107010d6ece21a85f1be4007bb","subject":"Merge \"Bump mypy version to 0.960\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2e6c0dfcecf18b107010d6ece21a85f1be4007bb"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2022-08-29 08:07:51.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2022-11-22 08:17:13.000000000","tz":-300},"subject":"Fujitsu Driver: Add Timeout function","message":"Fujitsu Driver: Add Timeout function\n\nWhen executing some functions, added the timeout for them.\n\nfollowing constants added\n* fujitsu_set_func_timeout\n* fujitsu_query_func_timeout\n* fujitsu_eternus_req_timeout\n* fujitsu_timeout_retry_times\n\nChange-Id: Ia36b8892e88453716f68956c44daeb14d8854912\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8faaebef1850c7e3f3bc0ee483bb859213eb9f29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8faaebef1850c7e3f3bc0ee483bb859213eb9f29"}]},"branch":"refs/heads/master"},"9ff369bb5b39af90647366552dc165e477f606f1":{"kind":"REWORK","_number":6,"created":"2023-04-06 07:24:41.000000000","uploader":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"ref":"refs/changes/54/854954/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/54/854954/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/854954/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/854954/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/54/854954/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/54/854954/6"}}},"commit":{"parents":[{"commit":"2e6c0dfcecf18b107010d6ece21a85f1be4007bb","subject":"Merge \"Bump mypy version to 0.960\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2e6c0dfcecf18b107010d6ece21a85f1be4007bb"}]}],"author":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2022-08-29 08:07:51.000000000","tz":-240},"committer":{"name":"inori","email":"xuq.fnstxz@fujitsu.com","date":"2023-04-06 07:29:12.000000000","tz":-240},"subject":"Fujitsu Driver: Add Timeout function","message":"Fujitsu Driver: Add Timeout function\n\nWhen executing some functions, added the timeout for them.\n\nfollowing constants added\n* fujitsu_set_func_timeout\n* fujitsu_query_func_timeout\n* fujitsu_eternus_req_timeout\n* fujitsu_timeout_retry_times\n\nChange-Id: Ia36b8892e88453716f68956c44daeb14d8854912\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ff369bb5b39af90647366552dc165e477f606f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9ff369bb5b39af90647366552dc165e477f606f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
