)]}'
{"id":"openstack%2Fcinder~859194","triplet_id":"openstack%2Fcinder~master~I14ef819f8b45b14c2453641cae5cb1b3bdb150d8","project":"openstack/cinder","branch":"master","attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-09-24 14:57:08.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I14ef819f8b45b14c2453641cae5cb1b3bdb150d8","subject":"WIP: Add readonly support in tgt target","status":"NEW","created":"2022-09-24 14:19:04.000000000","updated":"2022-09-30 00:57:07.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"972ef563e4ccece4926bdbd90f87a0e76d5181f9","_number":859194,"virtual_id_number":859194,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-24 17:25:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-24 14:39:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-24 14:47:15.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-24 14:57:08.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2022-09-24 15:14:03.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-09-24 15:24:14.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-09-24 15:30:56.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-24 17:09:52.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-09-24 17:25:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-24 17:28:20.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-09-24 17:32:06.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-09-24 17:48:13.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-24 19:47:00.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-09-25 04:46:03.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-09-26 19:24:56.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-30 00:57:07.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"}],"messages":[{"id":"9052f0833b4bb804892b874762d4260e7b70e489","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-09-24 14:19:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c4863f10aac3f51838f8c98406a8ee44ec7c9d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-24 14:39:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b4a072439604e31b339d902712a1e3e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/85d4e3b7992b4ab4b87b1a5fd69b1fdd : FAILURE in 17m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6523ceca349d4cba9709d5650cecc878 : FAILURE in 17m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a552cd3c2e840ebc1b34cc6c78a8f7ff2d267013","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-09-24 14:47:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1eb3f25064e8497b87713441e67e4f2a : RETRY_LIMIT in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"0b70c5a96ec1d57b7f6b06b2694bf03915d2b650","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-09-24 14:57:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/4ea1cffa364e4f81a9b759ff145c5630 : RETRY_LIMIT in 46s","accounts_in_message":[],"_revision_number":1},{"id":"ce8a5217fc3a4c3b26eadc7713e3eba277dedbd4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-09-24 15:14:03.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/859194/1/iscsi/ : SUCCESS in 54m 21s\n- cinder-smb http://cloudbase-ci.com/cinder/859194/1/smb/ : FAILURE in 13s","accounts_in_message":[],"_revision_number":1},{"id":"639bea064537b04ff8f51c07dd9b0b56528aa539","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-09-24 15:24:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/94/859194/1/check/iscsi-driver-alletra-6k/de8c049/de8c049.html : FAILURE in 1h 03m 52s","accounts_in_message":[],"_revision_number":1},{"id":"31fa93e95bf584bd2a0cdde4b005de21f9a8bb45","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-24 15:30:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/859194 : FAILURE in 1h 10m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5896eaaf71fcea4b594c1abf6ea270d09a4f8203","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-24 15:48:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/859194 : SUCCESS in 1h 27m 02s","accounts_in_message":[],"_revision_number":1},{"id":"6e00acc00fb434a20b31d1df00350be6d4fc438e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-09-24 17:09:52.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1664029154845 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"7607aea3b06349f7e9f6961dc54d2287772d8abc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-24 17:25:40.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/776e6ef55f324821bb403bb917e0b4f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba64126a832d4f5ea6dc560c8347e8bf : FAILURE in 7m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cc0e20dd88849078ab4fb64a446c769 : FAILURE in 6m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aac202180d5a47ee810f0bd6f093161f : FAILURE in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d890d77f3a274ccaa6cb57b32b6d4ebf : FAILURE in 7m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/a9492f0bb24d4228b32eafd428acf9f5 : FAILURE in 51m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/50afb4f70ec444d3b26cb4fdb36a4ad4 : FAILURE in 1h 19m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ede0cd95eaee4703a41b96191043dc3b : FAILURE in 48m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/01b49e60940d48f4824eaca4d5506428 : FAILURE in 7m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3d3cdfd09e494a6cbb72eaaa34a03617 : SUCCESS in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c009fc7995bc49af9d741ecacc72c5e1 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c59abdd68ecb468e93e789922748397c : FAILURE in 5m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/66c8dab448ea4cd0965f502b1a975b4c : FAILURE in 6m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0f269df791fa420cb7d36c60fa27372e : FAILURE in 51m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eae3795c529d4852aaeb19b456c459f0 : FAILURE in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b84b671604134341af847debe0f7d00d : SUCCESS in 1h 13m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5973d8417f134000ab109fe697a7f57e : FAILURE in 1h 26m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a10b62ceb30c41ec8cd722a784080c86 : TIMED_OUT in 3h 04m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e50093e8bdcc46358c7403f290807758 : SUCCESS in 1h 07m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fa44f72938da4d7e8f663854946eac1c : FAILURE in 1h 36m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d034d6959f18478e9866873c9a698f53 : FAILURE in 30m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3ef42ea559664b58a48018492fbf84eb : SUCCESS in 1h 19m 05s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2ceb69b96b1d4fb69a94eab40aa34226 : FAILURE in 2h 02m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/213f18f470cc42dfa0a4b428d846feba : FAILURE in 49m 41s","accounts_in_message":[],"_revision_number":1},{"id":"6200bda39e3ec08fdb955bafb59f14af9d2e959e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-24 17:28:20.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/94/859194/1/check/DellEMC_Unity_FC/d5e02ad/DellEMC_Unity_FC/1849 : SUCCESS in 1h 19m 57s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/94/859194/1/check/DellEMC_Unity_iSCSI/11b74f5/DellEMC_Unity_iSCSI/1696 : FAILURE in 1h 53m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"507315c3cf5eb3d45b7bdb575c923f99870b397c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-09-24 17:32:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/94/859194/1/check/dsvm-tempest-cisco-zonemanager-job/6fc3307/ : FAILURE in 1h 09m 04s","accounts_in_message":[],"_revision_number":1},{"id":"48492a1abd33acb663e93ac456460942dc08a79c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-24 17:48:13.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-94-859194-1/ns5_nfs : SUCCESS in 3h 28m 37s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-94-859194-1/ns5_iscsi : SUCCESS in 3h 26m 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"c12e6fc1c817e4d1a458b39c4eff97ce116e4f53","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-24 19:47:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/94/859194/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/d58bb2b/ : SUCCESS in 1h 33m 36s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/94/859194/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/cce7717/ : SUCCESS in 1h 35m 00s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/94/859194/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/0b36c3c/ : SUCCESS in 1h 34m 37s","accounts_in_message":[],"_revision_number":1},{"id":"fd75f2a35ab713ba3987961b9ec9490e4e7597e5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-25 04:46:03.000000000","message":"Patch Set 1:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/678","accounts_in_message":[],"_revision_number":1},{"id":"cd4c6eb460615e60bf9d4069977c956f95124eea","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-09-26 19:24:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/94/859194/1/20220927005110/NEC-ISCSI : SUCCESS in 3h 33m 22s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/94/859194/1/20220927005110/NEC-FC : SUCCESS in 3h 33m 43s","accounts_in_message":[],"_revision_number":1},{"id":"972ef563e4ccece4926bdbd90f87a0e76d5181f9","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-09-30 00:57:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/94/859194/1/20220930042922/NEC-V-ISCSI : SUCCESS in 2h 28m 39s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/94/859194/1/20220930042922/NEC-V-FC : SUCCESS in 5h 27m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f62359c50ac4c4bb4883d0f787036c754096306","revisions":{"7f62359c50ac4c4bb4883d0f787036c754096306":{"kind":"REWORK","_number":1,"created":"2022-09-24 14:19:04.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/94/859194/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/94/859194/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/94/859194/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/94/859194/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/94/859194/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/94/859194/1"}}},"commit":{"parents":[{"commit":"75735a4ca05622c04af7440d7e19173e62ba814e","subject":"WIP: Add read only attachment support for lio target","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/75735a4ca05622c04af7440d7e19173e62ba814e"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-09-24 14:17:43.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-09-24 14:17:43.000000000","tz":330},"subject":"WIP: Add readonly support in tgt target","message":"WIP: Add readonly support in tgt target\n\nthe readonly flag allows read only attachments in tgt target.\n\nChange-Id: I14ef819f8b45b14c2453641cae5cb1b3bdb150d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f62359c50ac4c4bb4883d0f787036c754096306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7f62359c50ac4c4bb4883d0f787036c754096306"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
