)]}'
{"id":"openstack%2Fcinder~859203","triplet_id":"openstack%2Fcinder~master~I40f6eb1f104da4410d32410940824a88f7d1ec62","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-09-26 05:10:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I40f6eb1f104da4410d32410940824a88f7d1ec62","subject":"Fix release note for bug 1957073","status":"MERGED","created":"2022-09-26 00:49:25.000000000","updated":"2022-09-26 09:58:22.000000000","submitted":"2022-09-26 05:10:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859203","meta_rev_id":"a0124040e24925a4e55c3eecd28ed46cebebf066","_number":859203,"virtual_id_number":859203,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-26 05:10:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-26 04:50:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-26 04:51:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-26 00:55:29.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-09-26 00:59:06.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-09-26 01:07:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-26 01:53:14.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-09-26 01:56:57.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2022-09-26 02:05:15.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-26 02:17:21.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-09-26 02:27:09.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-09-26 02:32:01.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-09-26 02:33:17.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-26 02:33:18.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-09-26 02:40:54.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-09-26 02:47:39.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-09-26 04:13:34.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-09-26 04:17:55.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"9963e9b39d849a11a00a4a2ee276b08c23a806c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-26 00:49:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fdc28ddac6433d28a5391641f17d9ee7de8888e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-26 00:55:22.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"822be6ae4781b6d9e05bfb9feb1d7a237c2e3c23","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-26 00:55:29.000000000","message":"Patch Set 1:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-trigger/683/ : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"0f7c48e0fe2b3c2d06703f68e86ea309b892372c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-09-26 00:59:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k finger://openstackcifc-virtual-machine/9a25379aca15462eb4e0784ca9927606 : RETRY_LIMIT in 11s","accounts_in_message":[],"_revision_number":2},{"id":"ec707f7200b009ee144c3132b4bbce192e5013a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 01:07:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58a003e78cb943e688c48301e9035cec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caa9824bb679498faf63198efc5f136a : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fea3b8c41b544665ab323868ea3764d5 : SUCCESS in 10m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2717a1ea79ff497d8793eefb78a8c8ba : SUCCESS in 7m 16s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7a64a9c8e55e4bde9f45c8988e67be60 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":2},{"id":"f066e19ae0cfb3d419bc2a83a95d042288dc24fa","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-09-26 01:53:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/859203/2/iscsi/ : SUCCESS in 52m 34s\n- cinder-smb http://cloudbase-ci.com/cinder/859203/2/smb/ : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":2},{"id":"3badc4e6dd2996c5c6adf8c590405f9b5fdb2154","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-09-26 01:56:57.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/917f6d142db64e8da5029e63368b3c7c : FAILURE in 1h 00m 27s","accounts_in_message":[],"_revision_number":2},{"id":"4f5a9990f018968a12cb964d85010d943d38ffe1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-26 02:05:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/859203 : FAILURE in 1h 08m 18s","accounts_in_message":[],"_revision_number":2},{"id":"89acb1d1c5a9133ca7d23d681762dece9cf32473","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-09-26 02:17:21.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/03/859203/2/check/dsvm-tempest-cisco-zonemanager-job/2fca6ee/ : FAILURE in 1h 11m 26s","accounts_in_message":[],"_revision_number":2},{"id":"efd650e99192e2ff2ce15c70a087785b07c01c71","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-26 02:27:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/03/859203/2/upstream-check/cinder-tempest-plugin-ontap-fcp/c457a63/ : SUCCESS in 57m 58s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/03/859203/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/09b6063/ : SUCCESS in 1h 13m 32s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/03/859203/2/upstream-check/cinder-tempest-plugin-ontap-nfs/2a4ab13/ : SUCCESS in 44m 54s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/03/859203/2/upstream-check/cinder-tempest-plugin-solidfire-iscsi/7d3005a/ : FAILURE in 1h 14m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"b6cdb88018bcb7b9518e429775e20682f9205b6e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-26 02:29:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/859203 : SUCCESS in 1h 31m 57s","accounts_in_message":[],"_revision_number":2},{"id":"6db00847fc1bea47ed1308d8c88173739fd4a257","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-09-26 02:32:01.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1664153373495 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"e64ccfcddf21ec8c576b5d4fb97857c6fa517a9a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-09-26 02:33:17.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/03/859203/2/check/DellEMC_PowerFlex/ab4c8fa/DellEMC_PowerFlex/1363 : FAILURE in 1h 25m 38s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"3718fce6e3a6cf08ca142828cbe9f1fad51402d0","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-09-26 02:33:18.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/03/859203/2/check/DellEMC_SC_FC/05089ea/DellEMC_SC_FC/1016 : SUCCESS in 55m 23s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/03/859203/2/check/DellEMC_SC_iSCSI/e01ef71/DellEMC_SC_iSCSI/1000 : SUCCESS in 1h 25m 42s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"00318857fda3fb694247392d8a0ca6352ffb3b06","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-26 02:40:54.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/03/859203/2/check/DellEMC_Unity_FC/cd01557/DellEMC_Unity_FC/1852 : SUCCESS in 1h 09m 48s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/03/859203/2/check/DellEMC_Unity_iSCSI/bf09ea6/DellEMC_Unity_iSCSI/1699 : SUCCESS in 1h 45m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"5e26d3aaf1567f381803c68445c28191e865dfc8","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-09-26 02:47:39.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/03/859203/2/check/DellEMC_PowerStore_iSCSI/c5427e6/DellEMC_PowerStore_iSCSI/800 : SUCCESS in 1h 26m 42s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/03/859203/2/check/DellEMC_PowerStore_FC/ba8a891/DellEMC_PowerStore_FC/952 : SUCCESS in 1h 27m 36s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/03/859203/2/check/DellEMC_PowerStore_NFS/fd586b2/DellEMC_PowerStore_NFS/491 : SUCCESS in 48m 39s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/03/859203/2/check/DellEMC_PowerStore_NVMeTCP/b9ed9c7/DellEMC_PowerStore_NVMeTCP/362 : SUCCESS in 1h 40m 00s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"f788fa880383f628e7595350a8614e91426a622a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-26 03:28:36.000000000","message":"Patch Set 2:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/684","accounts_in_message":[],"_revision_number":2},{"id":"3f354fd18b93574c34e82ecfc876eb3117fceb5a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-09-26 04:12:42.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1664153733502 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"e9772bf4a5c2e4c32ffb9bb5c73012f8f7c11dcf","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-09-26 04:13:34.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d80b3b8003e6f1407b4f90af0af00c0580208a6f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-26 04:17:55.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-03-859203-1/ns5_nfs : SUCCESS in 3h 27m 49s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-03-859203-1/ns5_iscsi : SUCCESS in 3h 28m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"3c0f4c21d1843b6fab8ebcb0da6e05665f21d58e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-26 04:43:43.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-03-859203-2/ns5_nfs : SUCCESS in 3h 29m 11s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-03-859203-2/ns5_iscsi : SUCCESS in 3h 31m 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"f6a311f48b6b0e2cc0036c2693eec973fbd02884","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-09-26 04:50:31.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6caf06926d5b1d6e7259f440c3e129d32cd9a85","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-09-26 04:51:35.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"90aee23547365fd6858e6c429f01a05fc8f3b0db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 04:51:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a4ebaa0a3742c7f4143ddc25626c330cf0d49848","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 05:10:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/780653462979441b82c1e5a648059292\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f60b43c48fe644d89e4dcc2ff8e87fd0 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7fca7d101814d3b900542b6491a5227 : SUCCESS in 12m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e37c15a96b604932accc047fdb245a9e : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":2},{"id":"6e92c8a5d2199ce4bc888070efc3643d34d2e1f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 05:10:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c575d1175f26e4491538335604ae970a4517fb48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 05:12:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c05b36296db8435988cb221f3b8da45b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46f2dbd3e20d47e58956227219ea420c : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29e17c871f2f45a1884fdda3a90d57f2 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2},{"id":"a0124040e24925a4e55c3eecd28ed46cebebf066","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-26 09:58:22.000000000","message":"Patch Set 2:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/686 \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b4592d44da666ddeae99f3d7ac6de005641bfb78","revisions":{"ff537326cd9b1459f4788a4947364e1cc2e597c8":{"kind":"REWORK","_number":1,"created":"2022-09-26 00:49:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/859203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/859203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/859203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/859203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/859203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/859203/1"}}},"commit":{"parents":[{"commit":"ebd8df7bcc3282bd28211e094919db2c84f9de4a","subject":"Merge \"Clean up formatting in dplcommon.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ebd8df7bcc3282bd28211e094919db2c84f9de4a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-26 00:48:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-26 00:48:31.000000000","tz":540},"subject":"Fix release note for bug 1957073","message":"Fix release note for bug 1957073\n\nThis is follow-up of Icd9dad9ad7b3ad71b3962b078e5b94670ac41c87 and\nfixes the release note based on the feedback given before the change\nis merged.\n\nChange-Id: I40f6eb1f104da4410d32410940824a88f7d1ec62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ff537326cd9b1459f4788a4947364e1cc2e597c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ff537326cd9b1459f4788a4947364e1cc2e597c8"}]},"branch":"refs/heads/master"},"b4592d44da666ddeae99f3d7ac6de005641bfb78":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-26 00:55:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/859203/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/03/859203/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/859203/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/859203/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/03/859203/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/03/859203/2"}}},"commit":{"parents":[{"commit":"ebd8df7bcc3282bd28211e094919db2c84f9de4a","subject":"Merge \"Clean up formatting in dplcommon.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ebd8df7bcc3282bd28211e094919db2c84f9de4a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-26 00:48:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-26 00:54:50.000000000","tz":540},"subject":"Fix release note for bug 1957073","message":"Fix release note for bug 1957073\n\nThis is follow-up of Icd9dad9ad7b3ad71b3962b078e5b94670ac41c87 and\nfixes the release note based on the feedback given before the change\nis merged.\n\nThe original change was merged after Zed release was cut off so this\nis fixed in master first.\n\nChange-Id: I40f6eb1f104da4410d32410940824a88f7d1ec62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b4592d44da666ddeae99f3d7ac6de005641bfb78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b4592d44da666ddeae99f3d7ac6de005641bfb78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
