)]}'
{"id":"openstack%2Fcinder~859319","triplet_id":"openstack%2Fcinder~master~I1f0d7679549954107abbfcc84efa9c05bc638852","project":"openstack/cinder","branch":"master","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-05-18 18:31:22.000000000","reason":"Vote got outdated and was removed: Code-Review+2"},"35075":{"account":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"last_update":"2026-05-18 18:31:22.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2026-05-18 18:31:22.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-05-18 19:50:04.000000000","reason":"Someone else replied on the change"},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2026-05-18 18:31:22.000000000","reason":"Vote got outdated and was removed: Code-Review+2"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I1f0d7679549954107abbfcc84efa9c05bc638852","subject":"Move check_ssh_injection to ssh_utils","status":"NEW","created":"2022-09-26 16:53:49.000000000","updated":"2026-06-19 20:04:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"42a747ebb22b883f8303bef0aac739ced25b30e7","_number":859319,"virtual_id_number":859319,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-18 23:44:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-26 17:38:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-26 18:23:50.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-09-26 18:55:34.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-09-26 19:08:04.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-09-26 19:31:09.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-26 19:36:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-26 19:41:55.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-26 19:56:24.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2022-09-26 21:31:00.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-26 22:20:51.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-09-26 22:44:49.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-09-27 00:43:09.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-09-27 01:02:56.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-09-28 14:40:39.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-28 17:56:56.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-28 18:23:08.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-09-28 18:42:00.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-09-28 18:44:16.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-10-15 11:42:59.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-12-16 14:29:38.000000000","updated_by":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"reviewer":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"state":"REVIEWER"},{"updated":"2022-12-16 14:56:26.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2022-12-16 14:58:39.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-12-28 09:20:44.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-12-30 03:18:26.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-12 13:15:41.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-12-12 15:59:54.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-12-13 15:11:36.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-18 19:50:04.000000000","updated_by":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"reviewer":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"state":"CC"},{"updated":"2026-05-18 19:50:57.000000000","updated_by":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"reviewer":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"state":"CC"},{"updated":"2026-05-18 21:18:07.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-18 21:53:38.000000000","updated_by":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"reviewer":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"state":"CC"},{"updated":"2026-05-18 23:46:34.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-06-19 20:04:19.000000000","updated_by":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"reviewer":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"state":"CC"}],"messages":[{"id":"128681e3dbf3927083ad1a78bcc1e2a377e44159","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-09-26 16:53:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8562ded2aebca3172138ab5631db9d3a9403713d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 17:38:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ed4cbba3049426396568ad92eb49df8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cff48e46998c457e8aa95307ca546623 : SUCCESS in 22m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3b6c21bdb19141af8ab60d7dec45391a : SUCCESS in 22m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1bfea753b4f9878fe136c4ee539047d066c827a5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-09-26 18:23:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/859319/1/iscsi/ : SUCCESS in 1h 25m 17s\n- cinder-smb http://cloudbase-ci.com/cinder/859319/1/smb/ : SUCCESS in 1h 08m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0c29163614eb3c7ade4436971dd932ec7468c243","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-09-26 18:55:34.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1664211240483 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"bd71fff8e2d9c800cbf01b75b78765d044dfcdc5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-09-26 19:08:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/19/859319/1/check/iscsi-driver-alletra-6k/4b372c4/4b372c4.html : FAILURE in 53m 38s","accounts_in_message":[],"_revision_number":1},{"id":"4a873066b0fcd290093aa40f984956fb809f6526","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-09-26 19:31:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/62f86aabb0e746b59b6233242710bc77 : SUCCESS in 37m 49s","accounts_in_message":[],"_revision_number":1},{"id":"1da82c7fcf0649b06cfdee3c8a26c23e4a1e40ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 19:36:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8beaa561dec142f8aa6d8a723ab9216a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7571887982434e01a2de1c894125c90c : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06808b5a85ff4adab2bfc1d25cafa0d1 : SUCCESS in 11m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce33d16b2f594c5cbfb79a384315edf6 : SUCCESS in 13m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d8266e70f9d49b28084e741cc7ed8cb : FAILURE in 7m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/5aba272de4c344a980a21026940efe55 : SUCCESS in 1h 01m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bcb95d690b6b4319b3e9f1077f8da890 : SUCCESS in 1h 15m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84ccb7244001452cb960836fc7d7bde6 : SUCCESS in 59m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/46f9bdbca6b74800b4110adc541cc732 : FAILURE in 7m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/dba8231d089f475ea0cdc1f344078c1e : SUCCESS in 7m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f86ce75006c84b77bb8deea0163f885a : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b895202701b04453942f692d417d6452 : SUCCESS in 6m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0b9a25cef19d4c07b7547aebdca0cd7a : SUCCESS in 7m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2df9c7f35e834c1cbd98e9f58a3eb942 : SUCCESS in 39m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/06a6dafddd0b41edbbc3868ad9fb8963 : FAILURE in 8m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/922bc1c9e62a4109814f92c52be8ec0a : SUCCESS in 1h 15m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/950556703fec4ab1ae4c9b506e0e7329 : FAILURE in 1h 29m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/78a7c79ccce443d4b0929e64735c5a4e : SUCCESS in 1h 35m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/19a6de3c90d8480b8b5df8ad03c3f2ce : SUCCESS in 1h 06m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0aa5ea4067404a7daeeb53916bd4741f : SUCCESS in 1h 19m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/41e959a16a394364b88667758602a65d : SUCCESS in 44m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/90068e72ad6449cd83374b758ccd3e1d : SUCCESS in 1h 34m 44s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e994b291c91b4d09b74ecd88126b9505 : SUCCESS in 2h 35m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25833d880a8446419af602718ae91d76 : SUCCESS in 54m 50s","accounts_in_message":[],"_revision_number":1},{"id":"afa752f9d750829d6bc2cba5ea4ae6c152b6ceb3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-26 19:41:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/859319 : FAILURE in 58m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5fec8f2e13b141245bdcdfef15f1f5a86f2eae5d","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-09-26 19:56:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/de35ba9d1bde47edace386147b3265fd : FAILURE in 55m 47s","accounts_in_message":[],"_revision_number":1},{"id":"8e6e27f7b14dfdc63ec41ba22c20aa78cbc17503","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-26 20:13:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/859319 : FAILURE in 1h 22m 23s","accounts_in_message":[],"_revision_number":1},{"id":"3ccdfb2f6ed6f59fe0939bffe0ee03fece41a0c4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-26 21:31:00.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-19-859319-1/ns5_nfs : SUCCESS in 3h 28m 58s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-19-859319-1/ns5_iscsi : SUCCESS in 3h 26m 36s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"da46e7d27fb0c7dc139be8b8ea4e4ad7eaec38dc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-09-26 22:20:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/19/859319/1/check/dsvm-tempest-cisco-zonemanager-job/f343edc/ : FAILURE in 1h 12m 38s","accounts_in_message":[],"_revision_number":1},{"id":"30bc693af93d57f0fd6b9d2ec14f861a3b8aa961","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-26 22:44:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/19/859319/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/826aefc/ : FAILURE in 28m 11s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/19/859319/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/9feee53/ : FAILURE in 27m 52s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/19/859319/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/5b02b2e/ : FAILURE in 28m 33s","accounts_in_message":[],"_revision_number":1},{"id":"17513ab22eeb600102dbdb3e2330815462f61858","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-27 00:43:09.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/19/859319/1/check/DellEMC_Unity_FC/50767ac/DellEMC_Unity_FC/1861 : SUCCESS in 1h 11m 02s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/19/859319/1/check/DellEMC_Unity_iSCSI/57b6ccc/DellEMC_Unity_iSCSI/1708 : SUCCESS in 1h 48m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"595a4274c2b662a6077ea3b14526051e27bf6fe9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-27 01:02:56.000000000","message":"Patch Set 1:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/692","accounts_in_message":[],"_revision_number":1},{"id":"64a399ff2b8a31ccac727ae6802173387331a8c7","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-09-27 12:51:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2426daf28ad9084cc4da3162707dadf0ecbcef0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-09-27 13:07:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5d1826483ad95b1c807a113097b3c4ee4ab960b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 13:36:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/558845f1bb464a788247eaa423a454a1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/27653b98908c40e0b79dbb856771b365 : SUCCESS in 13m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e9e74373ac124646bc55d11580445b15 : SUCCESS in 26m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1f4e30590a229477de27015165d230d9c82c7f79","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-09-27 14:02:01.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/19/859319/2/check/iscsi-driver-alletra-6k/1a8b2c6/1a8b2c6.html : FAILURE in 52m 44s","accounts_in_message":[],"_revision_number":2},{"id":"65cb848e4040ddefb27b92857ec981038abc67b6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-27 14:11:52.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/859319 : FAILURE in 1h 02m 13s","accounts_in_message":[],"_revision_number":2},{"id":"9dc0b847390bc7f654baa270a0d3c91304625e8a","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-09-27 14:15:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/e8fc9dd63a864eaebd21febf10dda654 : FAILURE in 1h 07m 10s","accounts_in_message":[],"_revision_number":2},{"id":"2f044780f7f4341189bbd52cb7509a89ed3a1220","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-27 14:15:52.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/859319 : FAILURE in 1h 06m 34s","accounts_in_message":[],"_revision_number":2},{"id":"25ce798360697db14af419793e3ccc84e57162b5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-09-27 14:17:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/859319/2/iscsi/ : SUCCESS in 1h 09m 41s\n- cinder-smb http://cloudbase-ci.com/cinder/859319/2/smb/ : SUCCESS in 58m 28s","accounts_in_message":[],"_revision_number":2},{"id":"79a433a0d1738a4b62c18707053d4c18043419c4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-09-27 14:38:44.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/19/859319/2/check/dsvm-tempest-cisco-zonemanager-job/8a0092e/ : FAILURE in 1h 13m 58s","accounts_in_message":[],"_revision_number":2},{"id":"b096070231be1cfef5fa646fcbe1f046d65f50c2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-27 14:50:26.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/19/859319/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/c3f49b2/ : SUCCESS in 1h 39m 32s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/19/859319/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3d72002/ : SUCCESS in 1h 35m 24s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/19/859319/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/a66d1aa/ : SUCCESS in 1h 42m 12s","accounts_in_message":[],"_revision_number":2},{"id":"baa9aaa84dbbc2b46a07372688c6b5386dc36182","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-09-27 14:54:11.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1664284033075 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"0c58d5290865c7c4e028cfba1dfe612203ab8934","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-27 15:07:35.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/19/859319/2/check/DellEMC_Unity_FC/356e934/DellEMC_Unity_FC/1863 : FAILURE in 1h 18m 50s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/19/859319/2/check/DellEMC_Unity_iSCSI/7a1cbdc/DellEMC_Unity_iSCSI/1710 : SUCCESS in 2h 00m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"c2fb5ee493b2282dd6dece56594551db3bafc3bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 15:22:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e98f020dffe42df95c6b4a4ae442d41\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7fdeeb6ae094584817dcf82993af642 : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44d1bcb8ff794884a3dc5563fc873eb0 : SUCCESS in 9m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2cf13af8ce9e42b4b5da9915fc40ba7f : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3be9d58e1a59415e92bd3fed0898f36a : SUCCESS in 12m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/5f11346554e44584b0661ce0f664b141 : SUCCESS in 54m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d202d97b5ef5476db79ec8878a638548 : SUCCESS in 1h 28m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb72b84d8880496db3ee6e72e05246e7 : SUCCESS in 55m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/aa32823f2ae848368c5f2a3ac683367a : SUCCESS in 15m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/94c31b7f40ca40e6bed174ba8924dd58 : SUCCESS in 5m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/17c59405893049ea8a7c4f9fc519f105 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c1f56e9041754fd8b9a29771301bf093 : SUCCESS in 6m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6a966cdc89e8462c9e354150a8e61010 : SUCCESS in 6m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6626c1d6fae54f58a4309fe5ed598688 : SUCCESS in 30m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dfd6d910fe4f461f9bca13b44e96dba8 : FAILURE in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/669c403a0e2c4772abd783025a606c9b : SUCCESS in 1h 13m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5a785d9fb0f648d8b170ae1643d1dd84 : FAILURE in 1h 24m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6261ffe5d9814d32aca6ff1a509dbc20 : SUCCESS in 1h 34m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3c84ea500121469db4b6f67bc4d57517 : SUCCESS in 1h 08m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0935a635b57044dd9a7c2a98a102b722 : SUCCESS in 1h 37m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4c1e3ced1db4457fa77d71676864089b : SUCCESS in 44m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/37646ab731ce45499f73531e36173a0e : SUCCESS in 1h 35m 44s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0f775f647abc41aa924e48652b8f5b4c : FAILURE in 2h 09m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fddc26a186a74872b12d62dc506d8561 : SUCCESS in 51m 40s","accounts_in_message":[],"_revision_number":2},{"id":"5f4817eceb89a37c5e8c07c0271364f92785fbbe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-27 15:50:35.000000000","message":"Patch Set 2:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/696","accounts_in_message":[],"_revision_number":2},{"id":"ca15d2f72a33f8cf27b5a134dd6b00543f23a8ba","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-27 16:38:09.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-19-859319-2/ns5_nfs : SUCCESS in 3h 30m 30s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-19-859319-2/ns5_iscsi : SUCCESS in 3h 30m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"08110b2a376e80c0586e3ffd910813e3b971535b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-09-28 14:40:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/19/859319/2/20220928200523/NEC-FC : SUCCESS in 3h 31m 42s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/19/859319/2/20220928200523/NEC-ISCSI : SUCCESS in 3h 35m 12s","accounts_in_message":[],"_revision_number":2},{"id":"d826209cde1cfc3909474212e6bcfadccfaabab4","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-09-28 14:58:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c6aecb3ec5dd3507736d5816647e8062f143add9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 15:25:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f5092d72bb14450bc9e903572151cc5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4161dac398e74e17876f6a59b35d98c3 : SUCCESS in 19m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/060928e62bb345378f6d162598fbf9a1 : SUCCESS in 19m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7958308446b2a51cf5b6be97fe0a0e290c0981ef","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-09-28 15:56:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/8d5994080e5c4103951036e6fdcf1925 : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":2},{"id":"29408030585c52a247ed21ba82b8ed90bcdd5846","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-09-28 15:59:03.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/859319/2/iscsi/ : SUCCESS in 56m 41s\n- cinder-smb http://cloudbase-ci.com/cinder/859319/2/smb/ : FAILURE in 59m 17s","accounts_in_message":[],"_revision_number":2},{"id":"6833d3e9c2a72e602e57b65c3aac1171991b0977","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-09-28 16:22:31.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/19/859319/2/check/dsvm-tempest-cisco-zonemanager-job/d6bcad8/ : FAILURE in 1h 12m 38s","accounts_in_message":[],"_revision_number":2},{"id":"e821c8d9f10d8594d4c43edb0687d6ef20b12609","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 16:59:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4a4f22365794d0f9066eef8ea0e2ddb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9defadd1cc44a74a1c8075a07d1d90b : SUCCESS in 7m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c9e02e642b64adbbdfc0c05cd147b08 : SUCCESS in 11m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3b262d417c2442499ed3ad507656218 : SUCCESS in 12m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0cb6a8d972b48f5bce837337bb407b0 : SUCCESS in 12m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/b82b60cd13484972ae31ed719dc0e064 : SUCCESS in 1h 12m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5556adacdea945b7b32664ae9185fc1a : SUCCESS in 1h 29m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3dfe3b9bf8034a4c98061ff5d72261d1 : SUCCESS in 58m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c013dd37c8c0422eafdd5d00b0515896 : SUCCESS in 15m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e22f68b69ed248faab94f9df1348ee8b : SUCCESS in 5m 40s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cf5eb7153ea44166874f8cdfb68b8358 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ec55bd3526df43f4b1b288f377ee6772 : SUCCESS in 6m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3e923e96273248738efe7dc1dfdf704b : SUCCESS in 6m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4d43b93c2b094487ba5afbb96eadade9 : SUCCESS in 49m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8f70003e96d84d6a80043e54286fa2c1 : FAILURE in 6m 50s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3710bafef486487fa9c561145af308f0 : SUCCESS in 1h 20m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4bb25b91a8c847ceb92f224949e5b361 : FAILURE in 1h 37m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9b76477235a54c29877375288b293000 : SUCCESS in 1h 47m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fdd65aa4687c4d03baa60ed21cdd3480 : SUCCESS in 1h 14m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3cd3ede44f434ec696c4ad6831e52358 : SUCCESS in 1h 25m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f4469e8054ec4feeba4afce0df3fda4f : SUCCESS in 34m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d45e54f9ed244d45915d50100308ade1 : SUCCESS in 1h 29m 58s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e4c1699068644b029500d1f11dd20d59 : SUCCESS in 1h 58m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7895810d5b54655a95d4f618f9e84d2 : SUCCESS in 50m 57s","accounts_in_message":[],"_revision_number":2},{"id":"db0b01cc0b7b8035bb3f13fcfdf94480bbdedf6b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-09-28 17:56:56.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/19/859319/2/check/DellEMC_PowerFlex/4c0b84d/DellEMC_PowerFlex/1382 : SUCCESS in 56m 37s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"44cffa5b13af930cf831ef6585d7ff029418df49","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-09-28 18:23:08.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/19/859319/2/check/DellEMC_SC_FC/5c69ca8/DellEMC_SC_FC/1028 : SUCCESS in 58m 19s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/19/859319/2/check/DellEMC_SC_iSCSI/a03f54b/DellEMC_SC_iSCSI/1012 : SUCCESS in 1h 22m 54s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"a57f7ce5cb9aef5386c14712072001703d89a737","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-09-28 18:42:00.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/19/859319/2/check/DellEMC_PowerStore_iSCSI/d22c1e5/DellEMC_PowerStore_iSCSI/812 : SUCCESS in 1h 29m 14s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/19/859319/2/check/DellEMC_PowerStore_FC/7caae73/DellEMC_PowerStore_FC/964 : SUCCESS in 1h 02m 51s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/19/859319/2/check/DellEMC_PowerStore_NFS/aa8d924/DellEMC_PowerStore_NFS/503 : SUCCESS in 40m 39s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/19/859319/2/check/DellEMC_PowerStore_NVMeTCP/cc4a0b3/DellEMC_PowerStore_NVMeTCP/374 : SUCCESS in 1h 41m 42s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"1f7eb75e874fd610cd173117638270775769605b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-28 18:44:16.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/859319/2/upstream-check/cinder-tempest-plugin-ontap-fcp/e26e417/ : SUCCESS in 1h 01m 35s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/859319/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/901b1a6/ : SUCCESS in 1h 22m 17s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/859319/2/upstream-check/cinder-tempest-plugin-ontap-nfs/a969ba0/ : SUCCESS in 42m 38s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/859319/2/upstream-check/cinder-tempest-plugin-solidfire-iscsi/85942c8/ : FAILURE in 1h 43m 05s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"d61dab09e3467eff2f7b15faba0705f3189dd943","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-28 18:55:50.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/19/859319/2/check/DellEMC_Unity_FC/2e3d1f6/DellEMC_Unity_FC/1871 : SUCCESS in 1h 14m 01s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/19/859319/2/check/DellEMC_Unity_iSCSI/791bc2c/DellEMC_Unity_iSCSI/1718 : SUCCESS in 1h 55m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"773f113e1bd14b8d19c86a6b8c09bbd41da3cfaa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-28 22:09:14.000000000","message":"Patch Set 2:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/705","accounts_in_message":[],"_revision_number":2},{"id":"4d9f22b350e0845bb44bc50917a7bb52a8d85286","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-09-29 14:07:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1e4d0b56f34f68dcab1ebcb47e19427d0c191f9a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-10-15 11:42:59.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/19/859319/2/20221015145648/NEC-V-ISCSI : SUCCESS in 2h 20m 21s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/19/859319/2/20221015145648/NEC-V-FC : SUCCESS in 5h 46m 07s","accounts_in_message":[],"_revision_number":2},{"id":"145e4c36d59350ac52eae87d785778fdf94ebcec","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2022-12-16 14:29:38.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7c169c1ebfb7348d5b553963e9b0afddded57d7c","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-12-16 14:56:26.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d62fd68ea463fe0c721c2f80b7fdbb63009dea40","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-12-16 14:58:39.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e99884be94662229b7d829551fdab3e3370fbae8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-12-16 14:58:54.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1669a6ec3ff4e1d0c86cc658dc00654bcf427c94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-12-21 17:04:03.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"47ef638bca018333d83b90bbf51a807b989671c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 17:29:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/499cef866c714a19a6fadd62947d697e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb5ba94bbb1d4897a87604a2bcb4452a : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/43f25c9c359d4e4d81d6735d4151feb7 : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/931ab7d0b6e54642a64aba0a986c776b : SUCCESS in 20m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"262493cee44fd9158b77cade7ab96ccc9d4d0f4f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-12-21 18:26:39.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/3977e606c8c74ba0922912ff0a995433 : RETRY_LIMIT in 51s\n- pure-devstack-cinder-tempest-fc-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/3ceddc1783334440a6ebee95c0bed217 : RETRY_LIMIT in 50s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/dc780b6bb47a4157872d58bf0a2f79ea : RETRY_LIMIT in 50s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/7240a6b4d0ca44808bf4c4cd8d530691 : RETRY_LIMIT in 50s","accounts_in_message":[],"_revision_number":3},{"id":"0f77d6a77c26edf3155f7efe81ac4a13c46453b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 19:22:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c49ab474e259414fa870fd5b1737f410\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9168ab65ef964b57b669e78c45930e30 : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5dcc2f5648a4c07b0cc650705fa6387 : SUCCESS in 11m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/338d01b22fce4d2ca9c528c72399b785 : SUCCESS in 10m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1861a38519c42bda195ee7dfce5a1a4 : SUCCESS in 12m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/8d83ddc9109b44fb9e11cd62a650343f : SUCCESS in 54m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/946aa246c46e4624b0f97166579c8cca : SUCCESS in 1h 05m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/48aaf401f7a1426e9fd4fedb91cf0fa5 : SUCCESS in 1h 28m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38b63109ab15400eb4248b813a016c5d : SUCCESS in 52m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9e2797b7e07442b1a12f86a587231441 : SUCCESS in 18m 10s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/688d804135d642c7ad17c8689ea4ed7c : SUCCESS in 13m 42s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0c14df7789a143db80f6ff4c427d592f : SUCCESS in 5m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a90020d46eae439bb3da3ab5f12699d2 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/968955f24da247539ce18c5dd808fd74 : SUCCESS in 6m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1dc8afb8b82c4db0a50c471f4d31ea09 : SUCCESS in 6m 16s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1f03838ee7f44a10b338792940c69819 : SUCCESS in 54m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/17307f5234b646d58710d4b776661098 : SUCCESS in 5m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/52489bd9643b4f969feed7d6357fa200 : SUCCESS in 1h 48m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c21ed3a64ba745c0b2888e9bfa110298 : SUCCESS in 2h 11m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2498a825287b4ee49a806be310d46e2c : SUCCESS in 1h 52m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b43b709e8b33433cafbd0559b5f0a671 : SUCCESS in 2h 07m 07s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/95ed697e70d842b7bb9256a2d04cc843 : SUCCESS in 1h 03m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/695792a4b49c447e9ff1d38d52a52793 : SUCCESS in 1h 19m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/10620502e1934718bf8ec2a92ec8e879 : SUCCESS in 48m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3999e0a4324244c3a93ccdb83ffab220 : SUCCESS in 1h 04m 17s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/788bcb49899d4433b90d8d0590242246 : SUCCESS in 1h 16m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4f692ac10d174108aa6f461b60d43adb : SUCCESS in 1h 21m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f481f07329e460084a270d6fe7f584a : SUCCESS in 1h 04m 31s","accounts_in_message":[],"_revision_number":3},{"id":"bb7c02eaf9f1b4d9daaef3abc165fbf4cef14e40","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-12-21 21:15:44.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/b7565cea0db1429b95a9e95e40214557\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/cbd963ef81fb40ca92e7d072bdb8c02e : FAILURE in 1h 22m 42s","accounts_in_message":[],"_revision_number":3},{"id":"4a482c434fe948652e53b5ab0165945789181bd2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-12-21 21:51:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/859319/3/upstream-check/cinder-tempest-plugin-ontap-fcp/7c124b5/ : SUCCESS in 2h 27m 47s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/859319/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/3486d74/ : SUCCESS in 50m 20s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/859319/3/upstream-check/cinder-tempest-plugin-ontap-nfs/c49877d/ : SUCCESS in 41m 27s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/859319/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/c33cf75/ : SUCCESS in 2h 05m 17s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"67525f59708b0c01858abf4031fcadbca1af5ccf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-12-21 23:04:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/309/.","accounts_in_message":[],"_revision_number":3},{"id":"b37033eda70d587e458657a00f479594c34aa400","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-12-22 01:54:31.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/19/859319/3/check/DellEMC_PowerFlex_3.6/56ba3d7/DellEMC_PowerFlex_3.6/1102 : FAILURE in 1h 25m 55s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/19/859319/3/check/DellEMC_PowerFlex_4.5/581d9e5/DellEMC_PowerFlex_4.5/79 : SUCCESS in 1h 52m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"b20a87bdb627e561f5a38080b678aeeafbd8b8ab","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-22 02:28:47.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/19/859319/3/check/DellEMC_Unity_FC/3cb6f7d/DellEMC_Unity_FC/1149 : FAILURE in 1h 16m 15s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/19/859319/3/check/DellEMC_Unity_iSCSI/82709ea/DellEMC_Unity_iSCSI/1161 : FAILURE in 2h 07m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"7208311fc29ba1e08584b086620ba5c1642010f8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-12-22 06:44:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/325/.","accounts_in_message":[],"_revision_number":3},{"id":"199ff4be10bb9316cad532494547d26eb693c593","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-22 21:59:57.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1b280f48991b4995b1c6142062488392 : SUCCESS in 1h 08m 00s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/4acee76463df491aa54887eff41cf330 : SUCCESS in 1h 10m 56s","accounts_in_message":[],"_revision_number":3},{"id":"92d2919e750675f7cf9dae59e7258ef6026b25e1","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-12-23 02:30:03.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/19/859319/3/20231223075351/NEC-V-ISCSI : SUCCESS in 3h 09m 15s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/19/859319/3/20231223075351/NEC-V-FC : SUCCESS in 3h 36m 09s","accounts_in_message":[],"_revision_number":3},{"id":"413890f64bb0ca9d72b0da37368ec4071d73dd0f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-12-23 15:34:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/19/859319/3/20231223202409/NEC-FC : SUCCESS in 3h 59m 02s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/19/859319/3/20231223202409/NEC-ISCSI : SUCCESS in 4h 10m 23s","accounts_in_message":[],"_revision_number":3},{"id":"16cd4188fc75549a15db11bd64ea43a270516a27","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-12-28 09:20:44.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ee6f86083d502d4ad719f374f6805c22f1f334ab","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-12-30 03:18:26.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/19/859319/3/ : FAILURE in 1h 20m 41s","accounts_in_message":[],"_revision_number":3},{"id":"32488f938fdc848dc372892c0037decb2568aeeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-12-12 13:13:50.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2d17c507bb1c330c261c0e92463222949358db71","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-12-12 13:15:41.000000000","message":"Patch Set 4:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/1081/","accounts_in_message":[],"_revision_number":4},{"id":"408bb5fc4fc0b8204f18fd968980093c5e01aee3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-12-12 13:16:32.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/859319/4/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/859319/4/fc.log : FAILURE (skipped)","accounts_in_message":[],"_revision_number":4},{"id":"daf6eebbcc073628cc428c1759b45268f524b615","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 13:37:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fb4d20e0fcc43218f610c2b11669291\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/666c6928b5d04c02ac06dbb3a66107ac : SUCCESS in 19m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c60d566d689442e5861e155f492c14a8 : FAILURE in 20m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"727402d7a49d0c28032c57095b7a989316ac6a04","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-12-12 15:05:21.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3078c7e3a6bc4f2680dddf8354d28239\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c7faf49995964813b6b144f3527c57b4 : SUCCESS in 43m 08s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f6195a97ee8f4f379baf208d4669f256 : FAILURE in 1h 43m 27s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/98b67d14c441437b8c8e173ee1002973 : FAILURE in 1h 44m 13s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/0bf7d1d9eed44b14b8974fb6fda5d38a : SUCCESS in 40m 31s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b698c82390fb47968b46bda864d14b09 : RETRY_LIMIT in 30s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ff4444c57c3e4a4398d0a000a77218d1 : NODE_FAILURE Node request 200-0000018707 failed in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dc9fc2fddada4104a97cd6aef352b9ab : RETRY_LIMIT in 38s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cceb9a16ccc24a31b42da3a294e4fc77 : RETRY_LIMIT in 46s (non-voting)\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/cinder source: thirdparty-ci-config/zuul.d/project.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-iscsi-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#69\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-iscsi-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-fc-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-fc-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#84\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for devstack\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#481\u003e matched\n              Collecting job variants for devstack-minimal\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#438\u003e matched\n                Collecting job variants for devstack-base\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:master}} source: openstack/devstack/.zuul.yaml@master#314\u003e matched\n                  Collecting job variants for openstack-multinode-fips\n                  Variant \u003cJob openstack-multinode-fips branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#412\u003e matched\n                    Collecting job variants for base\n                    Variant \u003cJob base branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#2\u003e matched\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#281\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#314\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#361\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#241\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#261\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#301\u003e did not match\n                Variant \u003cJob devstack-base branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#301\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#483\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#403\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#436\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#482\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#357\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#378\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#379\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#419\u003e did not match\n              Variant \u003cJob devstack-minimal branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#422\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2023.2}} source: openstack/devstack/.zuul.yaml@stable/2023.2#524\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.1}} source: openstack/devstack/.zuul.yaml@stable/2024.1#444\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:stable/2024.2}} source: openstack/devstack/.zuul.yaml@stable/2024.2#479\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/2023.1}} source: openstack/devstack/.zuul.yaml@unmaintained/2023.1#523\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/victoria}} source: openstack/devstack/.zuul.yaml@unmaintained/victoria#396\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/wallaby}} source: openstack/devstack/.zuul.yaml@unmaintained/wallaby#417\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/xena}} source: openstack/devstack/.zuul.yaml@unmaintained/xena#418\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/yoga}} source: openstack/devstack/.zuul.yaml@unmaintained/yoga#458\u003e did not match\n            Variant \u003cJob devstack branches: {MatchAny:{ImpliedBranchMatcher:unmaintained/zed}} source: openstack/devstack/.zuul.yaml@unmaintained/zed#461\u003e did not match\n      Pipeline variant \u003cJob pure-devstack-cinder-tempest-fc-aio-jammy branches: None source: thirdparty-ci-config/zuul.d/project.yaml@master#17\u003e matched\n    Freezing job pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Collecting job variants for pure-devstack-cinder-tempest-nvme-roce-aio-jammy\n      Variant \u003cJob pure-devstack-cinder-tempest-nvme-roce-aio-jammy branches: {MatchAny:{BranchMatcher:stable/2023.1},{BranchMatcher:stable/2023.2},{BranchMatcher:stable/2024.1},{BranchMatcher:stable/2024.2},{BranchMatcher:master}} source: thirdparty-ci-config/zuul.d/jobs.yaml@master#39\u003e matched\n        Collecting job variants for pure-devstack-cinder-tempest\n        Variant \u003cJob pure-devstack-cinder-tempest branches: None source: thirdparty-ci-config/zuul.d/jobs.yaml@master#163\u003e matched\n          Collecting job variants for devstack-tempest\n          Variant \u003cJob devstack-tempest branches: None source: openstack/tempest/zuul.d/base.yaml@master#1\u003e matched\n            Collecting job variants for dev... (truncated)","accounts_in_message":[],"_revision_number":4},{"id":"5637aea2bb80b620d3037af29bbd3371fdc9dfa5","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-12-12 15:59:54.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/859319 : FAILURE in 33m 56s","accounts_in_message":[],"_revision_number":4},{"id":"c0f78429df0ce07cdf50588c83807bc7df8670d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 16:08:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c78ffe200cc41049eae7ba7e9f1ba8d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efdb4c4e5e274d88b1794f78d9ed364a : SUCCESS in 8m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74df86f8b4be4c6794d059611680eee1 : SUCCESS in 10m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/89ca57c54e8d4de488cd76c3666dfde5 : SUCCESS in 12m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6280730d38f44c68821ece04f4c1418e : SUCCESS in 7m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/37514f9f47cc41f682b764a08fd48d28 : SUCCESS in 36m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f69d2f1f3240435eb511fb8dc481281d : SUCCESS in 33m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a2015fd8205349e3b19b526e3f997a47 : SUCCESS in 1h 58m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fabb16b4fb9943279675f64b63e2a8dc : SUCCESS in 1h 02m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/90cb231359864755bd20cf7a6409e8d3 : SUCCESS in 21m 32s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1f34743ce82f4d809286dfdeb721ae25 : SUCCESS in 6m 47s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/72169307d557418d9e49b349f07cb473 : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/786bcf560e6f418aabbb4cb3f2ce9284 : SUCCESS in 5m 39s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d69d4a96a3ae4a5fa30861d04d8d483a : SUCCESS in 6m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fab60acda7cc49f8ae9533c9bc6a70d3 : FAILURE in 17m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b7ab5171fdd7433987ea827e282bc9bd : SUCCESS in 11m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d5bd335d5d914971ab440c916fe10df3 : SUCCESS in 1h 48m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/48971d9ff9124d419d9bed2dd4aa40e4 : SUCCESS in 2h 48m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/85dc76bc0eb04ddb95c934dd625acd09 : SUCCESS in 2h 17m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/14535e7720634530a8446304d1c5e814 : SUCCESS in 2h 08m 13s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/042ba1a1403d4de19986910c7f91611f : FAILURE in 38m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bfc4240b50a54469b3a5a7f3d6b04c74 : SUCCESS in 1h 11m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6589ec43f038410486bb5a1e8e6fd5de : SUCCESS in 1h 50m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7c9fe9f7def94231aa3cc33bdb1ac57e : SUCCESS in 37m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/eeaa89e0462643c59d90e0e4d5f46cd5 : SUCCESS in 1h 35m 59s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/c0bf8d34ce9141ee8a75fbef00b34cd4 : SUCCESS in 1h 26m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8b1ff1dabcd84a269a06c2e8aa783964 : SUCCESS in 58m 51s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4685c699d354b5bbd9da5205aaab4fb : SUCCESS in 1h 45m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4ee3abd4e66468487c557b89dd98bb2 : SUCCESS in 51m 22s","accounts_in_message":[],"_revision_number":4},{"id":"b4d216277a30a9a7aa7edeaf49fb5ba0fab431d4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-12-12 16:08:42.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/859319 : FAILURE in 34m 22s","accounts_in_message":[],"_revision_number":4},{"id":"e3c1a7c4af0782dad72ba92785e2766ff5c35177","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-12-12 17:06:32.000000000","message":"Patch Set 4:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/99/.","accounts_in_message":[],"_revision_number":4},{"id":"6c525ec33eb17c1aab12b5941f42f7feeecd9063","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-12-12 18:53:29.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/859319 : FAILURE in 43m 05s","accounts_in_message":[],"_revision_number":4},{"id":"a4dd37b5628e58afab617b10455b429ef69feb48","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-12-12 23:18:39.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1675148eb4114c38b75a9af657199868 : SUCCESS in 1h 08m 39s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/dda66d4aa7a84ad280b62c806b1c4ec6 : SUCCESS in 1h 25m 13s","accounts_in_message":[],"_revision_number":4},{"id":"c55167c0449417e96517dc714ac11ead05144d52","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-12-13 01:19:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/98/.","accounts_in_message":[],"_revision_number":4},{"id":"dedba9b19c1a1105aa79c7ee5f1d9bb0adc0de26","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-13 08:16:06.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/19/859319/4/check/DellEMC_PowerFlex_3.6/68c3ced/DellEMC_PowerFlex_3.6/2571 : FAILURE in 2h 44m 20s\n- DellEMC_PowerFlex_4.x https://elab-os-logsrv.delllabs.net/19/859319/4/check/DellEMC_PowerFlex_4.x/216db61/DellEMC_PowerFlex_4.x/1410 : FAILURE in 1h 56m 04s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/19/859319/4/check/DellEMC_PowerFlex_4.x_NVMeTCP/239b943/DellEMC_PowerFlex_4.x_NVMeTCP/1287 : FAILURE in 4h 46m 44s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"9791f845f59d62a0a1bb9a3d706db1ada7058009","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2024-12-13 15:11:36.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\nhttp://localhost:9000/t/fujitsuCI/buildset/5370ffdb323545c1adcdbfc6ab538260\n\n- fujitsu-eternus-dx-fc http://localhost:9000/t/fujitsuCI/build/40e461823bd54fc8a3f81c781b5e3a06 : FAILURE in 2h 41m 53s\n- fujitsu-eternus-dx-iscsi http://localhost:9000/t/fujitsuCI/build/a7c39630950b4ab4b26a663b0502888f : FAILURE in 2h 33m 37s","accounts_in_message":[],"_revision_number":4},{"id":"fb232d0b6817f7ecfe4bf6057c54466434cc995b","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-12-17 06:47:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/19/859319/4/20241217112045/NEC-V-FC : SUCCESS in 4h 01m 48s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/19/859319/4/20241217112045/NEC-V-ISCSI : SUCCESS in 4h 26m 28s","accounts_in_message":[],"_revision_number":4},{"id":"d7d23c2e7a64a90d53a6620c32d5e66764efa13a","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-12-18 17:43:50.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/19/859319/4/20241218205056/NEC-ISCSI : SUCCESS in 5h 23m 59s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/19/859319/4/20241218205056/NEC-FC : SUCCESS in 5h 52m 51s","accounts_in_message":[],"_revision_number":4},{"id":"6a2fe875ea57d229ccaf453a641e1401a9041c3c","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-01-13 15:21:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cd5c031c520eec817f4b569ddc1d2702b2b8035b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 15:47:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef847176b03f418c9fa94822de4bb17c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c72e3c4f6a424288b4376b435982abd9 : SUCCESS in 18m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e917cfb60a40493d93ad71b7213c8518 : FAILURE in 22m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"96fb03e68f41ea69e3aa6c71328b3cc8d5b02828","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-01-13 17:15:37.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9257ceff11ef460f871d57229d3160da : SUCCESS in 1h 22m 35s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3a0a45d24d0b4cf3aaf4280ea43b9a79 : SUCCESS in 1h 18m 00s","accounts_in_message":[],"_revision_number":4},{"id":"e4abaa5b3085d964bf8e9721bb6ebd620d4c52f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 18:00:55.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3420ed34eadd4aa09b997cebaa187d5a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33f35ad7b55144468d2f580b4438910a : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04f65de160a9478d9fb0c66ace2b2b30 : SUCCESS in 11m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f4b38233044b4d4a8a01a6c7f301572f : SUCCESS in 11m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8b313a7c7774b6687a89d19fa7ed760 : SUCCESS in 7m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/b32872753e8349b0a4465afda384c893 : SUCCESS in 51m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a5bffd8c20d5443f950a93c64c77a760 : SUCCESS in 48m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/55d595719796495496f9e8290b6c0723 : SUCCESS in 1h 35m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea043ca3adcd44e3b351a41b83a428a7 : SUCCESS in 55m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8ac810a6f4ef4c3eab9091c0ada74557 : TIMED_OUT in 40m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5846f57df60c465a9cb9e4ef01d824f8 : SUCCESS in 7m 30s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ea0b780fe4af4e57bc7110064d75def8 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bed87419e4f34efca6137dfc6ebcfbbd : SUCCESS in 6m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ddea62c9189240e8994c81454dbdfd48 : SUCCESS in 6m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cdc5f2c27f8a4407806ec61e4a5c1ccb : FAILURE in 34m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5d619f795e404c589c583b1663dd6c8d : SUCCESS in 8m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4567091237004d28a0c087377e2e87ee : FAILURE in 1h 40m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7abb760232e1486cbc3ad72cee9b01ce : FAILURE in 2h 33m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2b9ebf6a9d8f47d4928e1697bcb90def : SUCCESS in 1h 39m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7d2a2651a975429abaf361e90476afef : SUCCESS in 2h 07m 32s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/004dc829efe54e1fa9435c448a16ff39 : SUCCESS in 28m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5839d7edbadf4c04bb404b278daa71f6 : SUCCESS in 1h 13m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3a18da9ca74444c786554dbd21cef4c2 : SUCCESS in 1h 23m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/466bba1fe2be403c99da1657815f2c96 : SUCCESS in 50m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/876fd0552c98410c9a90057d515623d4 : SUCCESS in 1h 21m 05s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7f9b8b58dfb34349a6d8c02156306422 : SUCCESS in 1h 55m 59s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bf8e146c2f914fb39786ab8b32ec8b6e : SUCCESS in 1h 28m 01s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87cf722ff086407facd8100e6967837d : SUCCESS in 1h 33m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bb21d9ef15944d29f5e40cf8b836e67 : SUCCESS in 49m 32s","accounts_in_message":[],"_revision_number":4},{"id":"2e8384fc3b9bc6da84894b8b140262eec8a09d72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2026-05-18 18:31:22.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"22eacb8af399480f0ca5c248cfc132d417864523","tag":"autogenerated:zuul:check","author":{"_account_id":38252,"name":"HPE AlletraMP NVMeTCP CI","display_name":"HPE AlletraMP NVMeTCP CI","email":"yogeshwari.gitte-ext@hpe.com","username":"yogeshwarigitte"},"date":"2026-05-18 19:50:04.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-nvmetcp-driver build/3bcb08ea37034cbc962978585e85f83e : FAILURE in 24m 49s\n\nWarning:\n  NVMe CI log: https://github.com/hpe-storage/hpe-nvmetcp-ci-logs/tree/master/new_logs/859319","accounts_in_message":[],"_revision_number":5},{"id":"3e3cf5af0b4b681723c8a44a4cecc13fdc0668ef","tag":"autogenerated:zuul:check","author":{"_account_id":39096,"name":"HPE AlletraMP FC CI","display_name":"HPE AlletraMP FC CI","email":"diganta.bora@hpe.com","username":"digantabora"},"date":"2026-05-18 19:50:57.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-fc-driver build/921f6eab08404069be79418b513aff6d : FAILURE in 25m 20s\n\nWarning:\n  FC CI log: https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/859319","accounts_in_message":[],"_revision_number":5},{"id":"8094bf27e8dc7c6aa904c38caacd1f48de1fae28","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2026-05-18 19:57:47.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm build/c9053b2ede584fdca23557722243fc4a : FAILURE in 28m 37s\n\nWarning:\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/859319\n  iSCSI CI log: https://github.com/hpe-storage/hpe_cinder_logs/blob/master/new_logs/859319","accounts_in_message":[],"_revision_number":5},{"id":"e7d8b16df315169c413ead65f3ae1133ece11f9b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-05-18 21:13:42.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest build/4b2c88e32fd84c5ea5105a1069e8dabc : SUCCESS in 54m 32s\n- cinder-storpool-tempest-iscsi-multipath build/776e377d8d2a42bbaec5b9940f20c68a : SUCCESS in 59m 07s","accounts_in_message":[],"_revision_number":5},{"id":"9eba06fed1ed9f8ff78b48446fcd941da81009d2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-18 21:18:07.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  cinder_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1f0d7679549954107abbfcc84efa9c05bc638852 .","accounts_in_message":[],"_revision_number":5},{"id":"cfad97d600280f10892b73ea2aa86a7c328e6ad2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-18 21:36:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4004/.","accounts_in_message":[],"_revision_number":5},{"id":"c7a7ac2228f881865e83d232b718575d083df7e3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-18 21:45:12.000000000","message":"Patch Set 5:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4180/.","accounts_in_message":[],"_revision_number":5},{"id":"4925ee24775b3a52ce94dff5a9d46aba97007d89","author":{"_account_id":38770,"name":"HPE 3PAR Storage FC CI","display_name":"HPE 3PAR Storage FC CI","email":"shubhangip@hpe.com","username":"arcusci"},"date":"2026-05-18 21:53:38.000000000","message":"Patch Set 5:\n\nBuild Failed: 3par-fc-driver: Logs: https://github.com/hpe-storage/hpe-3parfc-ci/tree/main/new_logs/859319 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"4aba392187f3fc8be98b0567c76bd7795584ca37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 23:44:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dcf01d3d6d441bfb39504f51d20bba3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b312db8f13f4443fa0b35a51be50ae5b : SUCCESS in 6m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df21aafc0a974b16acde98a21f3025ec : SUCCESS in 11m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f0c1f6cb2eef43e681c595b579a76c12 : SUCCESS in 10m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bc25c69f3917451e9ed2190aa11e58c8 : SUCCESS in 15m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec7358e74e6846e0abbed8d57b5a0066 : SUCCESS in 13m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/aa37bd9ca2554503bb92c22f11cd9528 : SUCCESS in 58m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/11876fa750db4b97a1a967041e8ca947 : SUCCESS in 1h 03m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f55aee4c23304ad6b72ba3ffc3627d59 : SUCCESS in 1h 41m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b49af08dc5614bdda87b6cf0d079b751 : SUCCESS in 1h 05m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/865852c5a1494efba58b1bdc148e9786 : SUCCESS in 8m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c1616020ec01434fb4d023a09b19ed3e : TIMED_OUT in 41m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/db5ddb4c7f6141da8c1157c3d42792bc : SUCCESS in 4m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/058ecd10d92d405ebf95493c9a423523 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b16e6412334e445a97d2e0aee477da46 : SUCCESS in 6m 34s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c8aded1667e44bb0b05611e73e2b5d8d : SUCCESS in 6m 21s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/c247292eb5d945ae8a75c17a865d66f9 : SUCCESS in 8m 51s (non-voting)\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b474aad4914a4a0893f4e2e697d28ed8 : SUCCESS in 38m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c43e62fcde66478eb527dacc2bcb1c20 : SUCCESS in 6m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5fe8ccb07c1a4742bdadca85a1626c6c : SUCCESS in 1h 48m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0cb873a3d778448e89cd24ab1d17e83a : FAILURE in 1h 31m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0d3ce16c0b7147ecaed127ea4d959052 : SUCCESS in 1h 51m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-fips-py311 https://zuul.opendev.org/t/openstack/build/e219506ae98a48bcbe71abb09540f3ea : SUCCESS in 1h 58m 23s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/6b1fe36570e34ebda8ee9f3dd5a9ee0f : SUCCESS in 40m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d7ba554cf6c647a3b7e4227e49e8a13d : SUCCESS in 1h 13m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/cbdecd14f7474e5982826c669c47298f : SUCCESS in 1h 44m 48s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e976bae0845e42bca514e2e1cbb98dbb : SUCCESS in 50m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b236644e62bf46dba742842ab0afb304 : SUCCESS in 1h 40m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fc85b228b6f49abbe1c5b4fde1d484d : SUCCESS in 1h 00m 29s","accounts_in_message":[],"_revision_number":5},{"id":"3a8797c8ea9b5304396cd9a550981825a38ef0e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 23:51:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c85e9bb40da4700b790826b1e5543a3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4a3b1902b15345dbb5503bb1b52f2067 : SUCCESS in 14m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/81c338f198c141c49add1bc11a80ef2e : SUCCESS in 13m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/926388c4fad64c90b2449da41fb7cf8d : SUCCESS in 37m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7468afc7e5953b380b4db51f35d83d13ccdc92d7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-18 23:56:06.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_05_2026_23_49_29\nLeave a comment with","accounts_in_message":[],"_revision_number":5},{"id":"42a747ebb22b883f8303bef0aac739ced25b30e7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38702,"name":"Ranjith","display_name":"Ranjith","email":"goldranjith@gmail.com","username":"goldranjith"},"date":"2026-06-19 20:04:19.000000000","message":"Patch Set 5:\n\nBuild Failed","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c7f68c8ca55cb92515618f8efeb5605fb352f653","revisions":{"4781427bcc9a3cc4ecc6d46e9f56372826e4b42a":{"kind":"REWORK","_number":1,"created":"2022-09-26 16:53:49.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/19/859319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/859319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/859319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/859319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/859319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/859319/1"}}},"commit":{"parents":[{"commit":"e53eac17ff4cd18408829d0d9c562e746d149f98","subject":"Move check_exclusive_options to volume_utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e53eac17ff4cd18408829d0d9c562e746d149f98"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-09-26 14:55:53.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-09-26 16:37:59.000000000","tz":-240},"subject":"Move check_ssh_injection to ssh_utils","message":"Move check_ssh_injection to ssh_utils\n\nEveryone calling this code is importing ssh_utils\nalready, so move this ssh-related method there and\nout of the general Cinder utils.\n\nThis eliminates a number of cinder.utils imports in\ndrivers altogether.\n\nChange-Id: I1f0d7679549954107abbfcc84efa9c05bc638852\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4781427bcc9a3cc4ecc6d46e9f56372826e4b42a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4781427bcc9a3cc4ecc6d46e9f56372826e4b42a"}]},"branch":"refs/heads/master"},"4d816818517097abb01f210a91905f050e5f49d1":{"kind":"REWORK","_number":2,"created":"2022-09-27 13:07:05.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/19/859319/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/859319/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/859319/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/859319/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/859319/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/859319/2"}}},"commit":{"parents":[{"commit":"e53eac17ff4cd18408829d0d9c562e746d149f98","subject":"Move check_exclusive_options to volume_utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e53eac17ff4cd18408829d0d9c562e746d149f98"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-09-26 14:55:53.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-09-27 13:04:25.000000000","tz":-240},"subject":"Move check_ssh_injection to ssh_utils","message":"Move check_ssh_injection to ssh_utils\n\nEveryone calling this code is importing ssh_utils\nalready, so move this ssh-related method there and\nout of the general Cinder utils.\n\nThis eliminates a number of cinder.utils imports in\ndrivers altogether.\n\nChange-Id: I1f0d7679549954107abbfcc84efa9c05bc638852\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d816818517097abb01f210a91905f050e5f49d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d816818517097abb01f210a91905f050e5f49d1"}]},"branch":"refs/heads/master"},"f36328444b2cf197703ff3c05f0784039aac2405":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-12-21 17:04:03.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/19/859319/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/859319/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/859319/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/859319/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/859319/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/859319/3"}}},"commit":{"parents":[{"commit":"ce37e60a2417feb2a696a2c4344e142089daf2bd","subject":"Move check_exclusive_options to volume_utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ce37e60a2417feb2a696a2c4344e142089daf2bd"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-09-26 14:55:53.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-12-21 17:04:03.000000000","tz":0},"subject":"Move check_ssh_injection to ssh_utils","message":"Move check_ssh_injection to ssh_utils\n\nEveryone calling this code is importing ssh_utils\nalready, so move this ssh-related method there and\nout of the general Cinder utils.\n\nThis eliminates a number of cinder.utils imports in\ndrivers altogether.\n\nChange-Id: I1f0d7679549954107abbfcc84efa9c05bc638852\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f36328444b2cf197703ff3c05f0784039aac2405"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f36328444b2cf197703ff3c05f0784039aac2405"}]},"branch":"refs/heads/master","description":"Rebase"},"04b4cc8b203eb58512b3f59f08a28d5f3722bc3e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-12-12 13:13:50.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/19/859319/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/859319/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/859319/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/859319/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/859319/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/859319/4"}}},"commit":{"parents":[{"commit":"1d51752f3a906d7f3d037906d247fde61c79c18f","subject":"Move check_exclusive_options to volume_utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1d51752f3a906d7f3d037906d247fde61c79c18f"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-09-26 14:55:53.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-12-12 13:13:50.000000000","tz":0},"subject":"Move check_ssh_injection to ssh_utils","message":"Move check_ssh_injection to ssh_utils\n\nEveryone calling this code is importing ssh_utils\nalready, so move this ssh-related method there and\nout of the general Cinder utils.\n\nThis eliminates a number of cinder.utils imports in\ndrivers altogether.\n\nChange-Id: I1f0d7679549954107abbfcc84efa9c05bc638852\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04b4cc8b203eb58512b3f59f08a28d5f3722bc3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/04b4cc8b203eb58512b3f59f08a28d5f3722bc3e"}]},"branch":"refs/heads/master","description":"Rebase"},"c7f68c8ca55cb92515618f8efeb5605fb352f653":{"kind":"REWORK","_number":5,"created":"2026-05-18 18:31:22.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/19/859319/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/19/859319/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/859319/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/859319/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/19/859319/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/19/859319/5"}}},"commit":{"parents":[{"commit":"ac5ed1d3bbbb8df812877315144a31c261eda237","subject":"Move check_exclusive_options to volume_utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac5ed1d3bbbb8df812877315144a31c261eda237"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-09-26 14:55:53.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2026-05-18 18:30:47.000000000","tz":-240},"subject":"Move check_ssh_injection to ssh_utils","message":"Move check_ssh_injection to ssh_utils\n\nEveryone calling this code is importing ssh_utils\nalready, so move this ssh-related method there and\nout of the general Cinder utils.\n\nThis eliminates a number of cinder.utils imports in\ndrivers altogether.\n\nChange-Id: I1f0d7679549954107abbfcc84efa9c05bc638852\nSigned-off-by: Eric Harney \u003ceharney@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c7f68c8ca55cb92515618f8efeb5605fb352f653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c7f68c8ca55cb92515618f8efeb5605fb352f653"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
