)]}'
{"id":"openstack%2Fcinder~859447","triplet_id":"openstack%2Fcinder~master~I5b58e9485654f707740175d5ff2a60594dff42f5","project":"openstack/cinder","branch":"master","topic":"antelope_version","attention_set":{},"removed_from_attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2022-09-28 09:09:38.000000000","reason":"\u003cGERRIT_ACCOUNT_13425\u003e replied on the change","reason_account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"}},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-10-06 22:22:41.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"35316":{"account":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"last_update":"2022-10-06 23:49:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5b58e9485654f707740175d5ff2a60594dff42f5","subject":"Pure Storage - bump version numbers for Antelope","status":"MERGED","created":"2022-09-27 15:30:36.000000000","updated":"2022-10-06 23:50:44.000000000","submitted":"2022-10-06 23:49:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859447-antelope_version","meta_rev_id":"06cade9bc41eef21a57d64215d16081ea2b87b1d","_number":859447,"virtual_id_number":859447,"owner":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-06 23:49:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"all":[{"value":1,"date":"2022-10-04 13:01:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2022-10-04 10:06:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":2,"date":"2022-10-06 22:22:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-10-06 09:16:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":2,"date":"2022-10-04 16:41:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":1,"date":"2022-10-06 22:22:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-27 15:59:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-27 16:32:32.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2022-09-27 16:33:03.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-09-27 16:34:51.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-09-27 16:42:20.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-27 16:49:58.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-09-27 17:12:11.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-09-27 17:12:32.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-09-27 17:21:18.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-09-27 17:58:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-27 18:27:45.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-09-27 18:58:48.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-28 06:31:18.000000000","updated_by":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2022-09-28 10:08:39.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-28 10:15:11.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-09-28 10:30:21.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-09-28 10:41:04.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-09-28 18:16:51.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-29 05:03:32.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2022-10-03 20:28:02.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-10-04 05:43:57.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-04 10:06:43.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2022-10-04 10:15:57.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-10-04 10:44:19.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-10-04 16:41:47.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2022-10-06 09:16:07.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"REVIEWER"}],"messages":[{"id":"be80be29954a7048fc51872f9ac8b4151e62d710","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"date":"2022-09-27 15:30:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"007452d809f6922c47c61c87c3ddd8f72dea3ef9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 15:59:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c7617782343499eaae630d4d9966a40\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15a02449a3d149b387daf7c6f69ce755 : SUCCESS in 25m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d70b606443514d2f813570fe51770b70 : SUCCESS in 25m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"191c4ed864713affe832065f16e0ad9a1184f6cc","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-09-27 16:32:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/e0c563789f614e64b0753448a640b240 : FAILURE in 1h 01m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7f21bbc3c9d178edd4a700a63c4df5597e65e048","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-09-27 16:33:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/859447/1/iscsi/ : SUCCESS in 53m 59s\n- cinder-smb http://cloudbase-ci.com/cinder/859447/1/smb/ : SUCCESS in 1h 01m 56s","accounts_in_message":[],"_revision_number":1},{"id":"db57150ade28e3aa3275d5219ed1af09588a4d6d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-09-27 16:34:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/47/859447/1/check/iscsi-driver-alletra-6k/8cc2e62/8cc2e62.html : FAILURE in 1h 03m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d3a714f5628026f4052d35f9f81fadd7ac4138c5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-27 16:42:20.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/859447 : FAILURE in 1h 10m 09s","accounts_in_message":[],"_revision_number":1},{"id":"88ec7124634f3c9bae49fc847964c536cfe23078","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-09-27 16:49:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/47/859447/1/check/dsvm-tempest-cisco-zonemanager-job/efe3c82/ : FAILURE in 1h 11m 24s","accounts_in_message":[],"_revision_number":1},{"id":"d6e53a54ca08c57c4db205736347433afe88036e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-09-27 17:00:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/859447 : FAILURE in 1h 28m 33s","accounts_in_message":[],"_revision_number":1},{"id":"51fc0f4f5ec99a554855e9c08a626edd35a7880f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-09-27 17:12:11.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1664292643337 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"5edb05870da6d2b8b46ebd39a72677d75a729c48","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-27 17:12:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/47/859447/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/3ce6592/ : SUCCESS in 1h 41m 03s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/47/859447/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/181a619/ : SUCCESS in 1h 38m 30s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/47/859447/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/bfbb6eb/ : SUCCESS in 1h 38m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e09f0086c31ab2bcf5616de1b24cf2bab575d80f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-27 17:21:18.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/47/859447/1/check/DellEMC_Unity_FC/9af4f26/DellEMC_Unity_FC/1864 : SUCCESS in 1h 05m 14s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/47/859447/1/check/DellEMC_Unity_iSCSI/264331b/DellEMC_Unity_iSCSI/1711 : SUCCESS in 1h 50m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"bc6afc39db1d92bbd7e73cfc5d9fb68e3e0d2087","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 17:58:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/02f25699c70a46fdbe8c0507fae5e574\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ae0cfc382e845349d84018d618dd467 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4e0406e6493466ca349f024050f7d33 : SUCCESS in 10m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58053d6a96424277a5edc222443b3e83 : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b0c3a22b4344762913201812f5ccdcf : SUCCESS in 12m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/b768955c89e74a84948b5d7cf1d39b85 : FAILURE in 1h 05m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e3b0ee0e2b16452483b4cadb6cbf2ef9 : SUCCESS in 1h 30m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3dadba6461d4d17902765b172525e2a : SUCCESS in 58m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0beedb836f7400485e8bb1a9472e081 : SUCCESS in 8m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a1ea6f43e2584ddd9b243eaf0d3f43ea : SUCCESS in 14m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c743944d09834d2e861b2cd551e8971b : SUCCESS in 5m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6da9269d93224aa7bf7ca78bd49e325d : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/195d07cca9b645ac9f0d6cf310f4a237 : SUCCESS in 6m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3980d5faeaba4179ae12e7bd9238813b : SUCCESS in 5m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5610c257d84a452f823d1b1ac76d2c09 : SUCCESS in 32m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/befdca1af7f24c70993ac7343a3f3f61 : FAILURE in 5m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5a321ff70e3946cdbe3267d72755629d : SUCCESS in 1h 08m 21s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bfec9bcdd58a4b6abc4f2647f41edbe0 : FAILURE in 1h 51m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f0abaa2c9cb24dc5b5854c9782db8255 : SUCCESS in 1h 49m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/38140bbc326a40f89e5113dabff6fdae : SUCCESS in 1h 06m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/aa1c7f2a183d4564a2a322a5857ddbe7 : SUCCESS in 1h 41m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1432efc40e5347be9021afa9c68d2ee6 : SUCCESS in 37m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a8dd0126aa1d4c14ae11aad635122b0b : SUCCESS in 1h 22m 32s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b1bf3625a5e249a18316376f6fb09ea7 : SUCCESS in 2h 21m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db66d35d28344a9cab4c75a1964b3037 : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":1},{"id":"87a84c52051a74328904fd7a82af1afbb23346a9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-27 18:27:45.000000000","message":"Patch Set 1:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/697","accounts_in_message":[],"_revision_number":1},{"id":"6f62d1427d48e26007f44bbeb03de880ba5c3269","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-27 18:58:48.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-47-859447-1/ns5_nfs : SUCCESS in 3h 27m 42s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-47-859447-1/ns5_iscsi : SUCCESS in 3h 24m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"89a1c34b073c148701e9dd83d9e163b9b1713bc4","author":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"date":"2022-09-28 06:37:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8fee5f307b1109a1ed4479863ed4a5c39282ae3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 07:04:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/757db7fba2d344cdbfbc1ac04c61558a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/35015e2d96c641159287b705d2cf11a5 : SUCCESS in 23m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fb63f1bcef564ed49b15637de255f89c : SUCCESS in 24m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e117b187d0cb2d8810cbaa59796d8de49e1e9dae","author":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"date":"2022-09-28 07:17:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4365c383d9c96979aec43bd6a1d9a34befcf07b8","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-09-28 07:40:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/3f2f97e266cc41e7931578870bb7b81c : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b2ba21146f7095a3ba37cc10999d42dfebf0f3a6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-09-28 07:41:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/859447/1/iscsi/ : SUCCESS in 1h 00m 53s\n- cinder-smb http://cloudbase-ci.com/cinder/859447/1/smb/ : SUCCESS in 1h 03m 24s","accounts_in_message":[],"_revision_number":1},{"id":"72a284e209f9b2065ea8d16d4d74d4fa554f965d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-09-28 08:01:35.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/47/859447/1/check/dsvm-tempest-cisco-zonemanager-job/0545b1e/ : SUCCESS in 1h 14m 00s","accounts_in_message":[],"_revision_number":1},{"id":"79bbf1518ec667821a6fe1f7496b36e9923fe291","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-28 08:17:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/47/859447/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/bcacafd/ : SUCCESS in 1h 38m 31s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/47/859447/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/440d099/ : SUCCESS in 1h 32m 43s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/47/859447/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/f351153/ : SUCCESS in 1h 25m 56s","accounts_in_message":[],"_revision_number":1},{"id":"2c449b3facb83b738fa48d75f42e3d28d0811a0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 09:07:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b9f022524724348af8429bfa349e8b2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff82bfd215514133ab1858a08102ea34 : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9af1b18a86594e39b6916f0a44934535 : SUCCESS in 10m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9a46aae2d5d4abb9421415918aab25c : SUCCESS in 10m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8f0128c1de54b27a42640bb77ebc43b : SUCCESS in 15m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/6899badfafc74520b265acf35211abf0 : SUCCESS in 1h 06m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/739f4a0785ce46568ad1d21d0b2ec39e : SUCCESS in 1h 15m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1940aa939fa744d692e935e5873767c0 : SUCCESS in 55m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be0df291fc0a404d92f3ab68f3f64220 : SUCCESS in 8m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/99045281467c48d7af0ac0cacedbfc88 : SUCCESS in 17m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7384ccf122a04092a3e7449634033a2d : SUCCESS in 5m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9a287584f6f741b899ad1f8f6f0ba6ba : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/159b5b0c0f9a41a8be330a8e509f372d : SUCCESS in 5m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/24476f3df3b0405f952079062a8dae82 : SUCCESS in 5m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/92dce6989ca34b0183add8f0ee0091d6 : SUCCESS in 31m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/82407aaac62a4b83964dbfdfcf32d4d0 : FAILURE in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6b543bff75214833b0642f345057f9b9 : SUCCESS in 1h 09m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/092c6aac47f84017ba05e4295cf0e6dc : FAILURE in 1h 27m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3ed89466013948e4b5ff769a361d8b88 : SUCCESS in 1h 17m 01s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/75a07a6f252441019769c6c1b095ee0a : SUCCESS in 1h 10m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/da0d42d78fb44e229e9e40c23e0dd73d : SUCCESS in 1h 07m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/dd99d27d68b4437793483e5d103b6eec : SUCCESS in 46m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ffbabc5af19648edbf3835e3ace25de5 : SUCCESS in 1h 24m 32s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/608fa54483034025829673120284ca98 : SUCCESS in 2h 23m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac77ccc774bd489c83bcce1fbca6bb4b : SUCCESS in 41m 06s","accounts_in_message":[],"_revision_number":1},{"id":"0d1ac160a3f9d0711e5c0e9c83bab0293c7ba788","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-09-28 09:09:38.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b2eacc7f551460a80d26a442c41338724089c55","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-09-28 09:13:49.000000000","message":"Patch Set 1:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/700","accounts_in_message":[],"_revision_number":1},{"id":"4af706e28d6c735f336fd0f03d0191f5c3c781ab","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-09-28 10:08:39.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/47/859447/1/check/DellEMC_PowerFlex/c7b8c7f/DellEMC_PowerFlex/1380 : SUCCESS in 1h 00m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"591897486bbd98dfe12084715fc774aa0d41016f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-28 10:15:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/859447/1/upstream-check/cinder-tempest-plugin-ontap-fcp/96b0505/ : SUCCESS in 1h 00m 28s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/859447/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/1bb583c/ : SUCCESS in 42m 38s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/859447/1/upstream-check/cinder-tempest-plugin-ontap-nfs/ea1fb9c/ : SUCCESS in 45m 52s\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/859447/1/upstream-check/cinder-tempest-plugin-solidfire-iscsi/372dfe8/ : FAILURE in 1h 06m 06s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"0ba96c4602e5f2f39ecab3af1ce7f4d8c4acbbbb","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-09-28 10:30:21.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/47/859447/1/check/DellEMC_SC_FC/c603d3e/DellEMC_SC_FC/1026 : SUCCESS in 53m 51s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/47/859447/1/check/DellEMC_SC_iSCSI/60785a6/DellEMC_SC_iSCSI/1010 : SUCCESS in 1h 22m 13s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"8d395ddc02f675b3f1d055c87aa9993e7ea0a766","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-09-28 10:41:04.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/47/859447/1/check/DellEMC_PowerStore_iSCSI/2e36009/DellEMC_PowerStore_iSCSI/810 : SUCCESS in 1h 22m 23s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/47/859447/1/check/DellEMC_PowerStore_FC/74df71d/DellEMC_PowerStore_FC/962 : SUCCESS in 1h 06m 36s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/47/859447/1/check/DellEMC_PowerStore_NFS/0423d82/DellEMC_PowerStore_NFS/501 : SUCCESS in 41m 59s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/47/859447/1/check/DellEMC_PowerStore_NVMeTCP/b642e3a/DellEMC_PowerStore_NVMeTCP/372 : SUCCESS in 1h 32m 52s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"7f27dae4f1502dd5e0d80a3e424ee7ed3b80ec76","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-28 11:45:58.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/47/859447/1/check/DellEMC_Unity_FC/51d0bd6/DellEMC_Unity_FC/1868 : SUCCESS in 1h 11m 42s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/47/859447/1/check/DellEMC_Unity_iSCSI/04164a1/DellEMC_Unity_iSCSI/1715 : SUCCESS in 1h 51m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"a086f382207c95bf725dcbfae3c7153b3934d930","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-09-28 18:16:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/47/859447/1/20220928234043/NEC-FC : SUCCESS in 3h 34m 35s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/47/859447/1/20220928234043/NEC-ISCSI : SUCCESS in 3h 36m 05s","accounts_in_message":[],"_revision_number":1},{"id":"91f7e9b9c6bd0fef7cfbff13968f281e0c7858c7","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-09-29 05:03:32.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f1ec286c188d456c233f5fb78b7db3de0c963bb6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-10-03 20:28:02.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"88cb493e0ad688f0fce21f3c22e9f72d5e8ca2bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"date":"2022-10-04 04:37:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1110211d01f3633ee7dae4aff61a10e59857f389","author":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"date":"2022-10-04 04:41:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b575244310769677e304b1b75ea1a667acce4c1a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 04:59:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f867fac64b742cbbfe4e723dff4352a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26ad8b40aab34b3ea2566f039363f2ec : SUCCESS in 15m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c6ef4fb898f548ed937e628048005030 : SUCCESS in 13m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7ffda840c4c404951f1ff06918e51f636c29ea3c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-10-04 05:16:20.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/859447/2/iscsi/ : FAILURE in 38m 01s\n- cinder-smb http://cloudbase-ci.com/cinder/859447/2/smb/ : FAILURE in 38m 01s","accounts_in_message":[],"_revision_number":2},{"id":"13e1e8932a35a47b4fa45286752e2b1ae09ee4e4","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-10-04 05:36:25.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/16ba6205322c431aa566e5041dd5d58f : SUCCESS in 57m 47s","accounts_in_message":[],"_revision_number":2},{"id":"ed639117c987431251845c828fbedf6afd5555b7","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-10-04 05:43:57.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/47/859447/2/ : FAILURE in 58m 14s","accounts_in_message":[],"_revision_number":2},{"id":"a2a82567094d2a9ab4b604c86ba54b32f2d5903a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-10-04 05:45:36.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/47/859447/2/check/iscsi-driver-alletra-6k/f60430d/f60430d.html : FAILURE in 1h 06m 16s","accounts_in_message":[],"_revision_number":2},{"id":"6471505484e7a5f3a1e27abb3f046d38134bb9ae","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-10-04 05:50:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/47/859447/2/check/dsvm-tempest-cisco-zonemanager-job/dbf7003/ : FAILURE in 58m 51s","accounts_in_message":[],"_revision_number":2},{"id":"8285437a0290e7f0a8c525dca3769a3de1924645","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-04 06:17:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/859447 : SUCCESS in 1h 38m 08s","accounts_in_message":[],"_revision_number":2},{"id":"5a0f323c41a6e4dd798c32f6770ece769b9095e7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-10-04 06:19:40.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1664858277371 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"84c537957a5a9827d0083f86f2881b8723eb1652","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-04 06:21:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/47/859447/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/6c85cff/ : SUCCESS in 1h 34m 14s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/47/859447/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/20b2fdc/ : SUCCESS in 1h 42m 28s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/47/859447/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/bd09d80/ : SUCCESS in 1h 29m 03s","accounts_in_message":[],"_revision_number":2},{"id":"e3c6ee0f9a38f979b44ae48f804d6012d76490f7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-04 06:43:48.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/47/859447/2/check/DellEMC_Unity_FC/aa74cf8/DellEMC_Unity_FC/1890 : SUCCESS in 1h 11m 22s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/47/859447/2/check/DellEMC_Unity_iSCSI/063acdb/DellEMC_Unity_iSCSI/1737 : SUCCESS in 2h 01m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"4c8aa981072c5b0b6c515981c66163c5c1e8ede4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 06:52:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8dfd2f63ec374cc2864f4eb77194fcc7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bff9b49add3a4bbb8e006b9a766c2ff3 : SUCCESS in 7m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfe3aadf70634c6a861a8807209942c8 : SUCCESS in 11m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d76305d476944c8993d9c2bf6a4b3849 : SUCCESS in 11m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a7e91dd5d2e44898a0117aa7414d50a : SUCCESS in 11m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/498ed533c23d4e9da8e1b2b3d9c4d1bf : SUCCESS in 42m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/621afb2eedf04e65a099451ea8785ee9 : SUCCESS in 1h 04m 45s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9eda621f57ff4bf58159fc6fa9b9be1e : SUCCESS in 1h 08m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/056931da549a41ed99d12cf56bfaa916 : SUCCESS in 1h 02m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d4cf2fd6c2ff443f9d4e607593fede20 : SUCCESS in 14m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0b70604ce5c54a1bbdfb9f63d995a680 : SUCCESS in 8m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/32d27f3bd3614fdf87eebc30bd43d34b : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/29798b2720a14eaea7c65b1ce59aa2c2 : SUCCESS in 10m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8538ebbdd5844dddbd66f8cd5d9e9de6 : SUCCESS in 5m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4c619ad231ee467c9f6b0873e381962f : SUCCESS in 45m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9f394c56c2f94f509920c6eb6232d998 : FAILURE in 5m 36s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/36e8e1d5ab734e86979a942173e48126 : FAILURE in 1h 25m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d7d9de22303841ed8f0d0a21928f84ae : FAILURE in 1h 17m 52s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3c8e3bdbc57e4990b06855cb03f14669 : FAILURE in 1h 51m 49s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/86aed6619049465196135aa3d452e2bc : FAILURE in 42m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c9313ae392624fbb9d201f022ea3527d : SUCCESS in 1h 29m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ca4ed5a57fb74cafb48d0b494682a91e : SUCCESS in 33m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bd04771f8551420f9dbab5f80ad6e249 : SUCCESS in 1h 29m 45s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/90e9144956014bfb8ee5d30ec95b7532 : SUCCESS in 2h 10m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a007469b5deb4eec85808e477fd644aa : SUCCESS in 44m 41s","accounts_in_message":[],"_revision_number":2},{"id":"9b0f77bb6a5967cd8dd931b0274f710595667319","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-10-04 07:33:30.000000000","message":"Patch Set 2:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/732","accounts_in_message":[],"_revision_number":2},{"id":"424349f310d900c6df3b40323d9fcd63659547ca","author":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"date":"2022-10-04 07:57:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"82ad45fce7894e0a6c6b45f2fa3988ec8920bbd5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-10-04 08:10:59.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-47-859447-2/ns5_nfs : SUCCESS in 3h 30m 59s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-47-859447-2/ns5_iscsi : SUCCESS in 3h 32m 44s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"d4a68deb8349dc39ad43bf3fcde631b2ba586518","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 08:29:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e9da5d5c1c944228e80d7d8a688ca9e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3ef0e3ea36b64c1d85982a92c3c3f846 : SUCCESS in 25m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ba44a12ecfc94423848a5461e6137900 : SUCCESS in 23m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"09ff20da1d31aea79c7d36077b8683128a0b8aea","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-10-04 09:08:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/859447/2/iscsi/ : SUCCESS in 1h 09m 49s\n- cinder-smb http://cloudbase-ci.com/cinder/859447/2/smb/ : SUCCESS in 1h 05m 41s","accounts_in_message":[],"_revision_number":2},{"id":"3b829662412606f071e0b123e8fd10447f13f7f1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-04 09:30:18.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"1df838cf3733d3eeea3eb6215564651fa86e6d7e","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-10-04 09:54:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/3198fa6abed84f0fb4633fdd769275f2 : FAILURE in 57m 45s","accounts_in_message":[],"_revision_number":2},{"id":"7b4669f40fca959ce77e3d178d576f5279c1def0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 09:58:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2619346e3c044b41a74cb7b3c0d46571\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46b63b41449e4cd9b5acf1ae6ffc277c : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0170bac4ca7c4e0cac424c4740ba64d2 : SUCCESS in 10m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9eac3868cc934b8e8404e4e3b70052c7 : SUCCESS in 12m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cda176db3994175984e6b7b0ef943bb : SUCCESS in 11m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/0c384361b0df4c5090bbd2748e7a4b5f : SUCCESS in 53m 55s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d3e77f637b244d95b9456fe011dd8aa9 : SUCCESS in 50m 11s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2f60af48974046eb9cf95f234da320ca : SUCCESS in 1h 22m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90488e86faf0472a845562d2b205a51d : SUCCESS in 42m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c679ce60e6b7427a85234f187ed6ec34 : SUCCESS in 13m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5e895dac762e4a84b7deaa35766273f5 : SUCCESS in 5m 42s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d7cea55edad34eab94c7be51106e0e6f : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1a4b13a3740c4dd9806c2f4824626fe9 : SUCCESS in 6m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c52429453b144d3b8bc11a93a9a7131e : SUCCESS in 9m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c954082778994d09a8031c6947b63249 : SUCCESS in 35m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/748fd21fc40f44758c3e779648c0aa39 : FAILURE in 8m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f888612109a942e6aa6570fafe1af3e7 : SUCCESS in 1h 16m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d54d90ff09f64baea26de1d3d464ef48 : FAILURE in 1h 29m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7c2957f66ef54d27a056ded2839aa5c9 : SUCCESS in 1h 39m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3665691abeac4377ad5a11a33b381f9b : SUCCESS in 1h 15m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b01a926675e6412b83dd496ffbd4d3ab : SUCCESS in 1h 22m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a84db0fa0e7049c180838d821cbdd347 : SUCCESS in 33m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c9da0f43fd294fe89355c63c709727c2 : SUCCESS in 1h 34m 03s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/35ee53facbe4496b9137fc89e3fe9177 : SUCCESS in 1h 50m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ca4b92e88a14d23b060623af4739b89 : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":2},{"id":"ee0413b3866792f261600c3f6c6a8cd2f9c516f7","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-10-04 10:06:43.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7f9d73a8f96a421c9b7a8f433fbeffd1f9cba422","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-10-04 10:13:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/47/859447/2/check/dsvm-tempest-cisco-zonemanager-job/13c091f/ : FAILURE in 59m 01s","accounts_in_message":[],"_revision_number":2},{"id":"9fcb644566b623d396122f653b095685b168c1f2","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-10-04 10:15:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-47-859447-2/1664877491 : SUCCESS in 17m 43s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"aa38c5eeddc3987a591782a3b293730b748aa158","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-10-04 10:44:19.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/859447/2/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/66b1e73/ : FAILURE in 43m 53s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"5299bf18f64db3cde08eb293fb3b5a18033c88c5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-10-04 11:03:59.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/859447/2/upstream-check/cinder-tempest-plugin-ontap-fcp/449aa88/ : SUCCESS in 1h 05m 00s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/859447/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/1ef6c82/ : SUCCESS in 43m 56s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/859447/2/upstream-check/cinder-tempest-plugin-ontap-nfs/d04f83d/ : FAILURE in 1h 02m 39s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"40dd9d16e549128233d8e35cec963c96d33b9a43","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-10-04 11:06:32.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/47/859447/2/check/DellEMC_PowerFlex/96186d5/DellEMC_PowerFlex/1394 : FAILURE in 1h 08m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"21b71ed3796c78a6ad66da3718f8cf34225634a1","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-10-04 11:18:35.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/47/859447/2/check/DellEMC_SC_FC/96959a8/DellEMC_SC_FC/1040 : SUCCESS in 54m 21s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/47/859447/2/check/DellEMC_SC_iSCSI/ef46bad/DellEMC_SC_iSCSI/1024 : SUCCESS in 1h 20m 06s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"ec24c06e2fcc931e8448e00d18db7ec644e34250","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-10-04 11:44:41.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/47/859447/2/check/DellEMC_PowerStore_iSCSI/4da66e0/DellEMC_PowerStore_iSCSI/824 : SUCCESS in 1h 20m 11s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/47/859447/2/check/DellEMC_PowerStore_FC/d5cbc1f/DellEMC_PowerStore_FC/976 : SUCCESS in 1h 04m 24s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/47/859447/2/check/DellEMC_PowerStore_NFS/a541043/DellEMC_PowerStore_NFS/515 : SUCCESS in 45m 15s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/47/859447/2/check/DellEMC_PowerStore_NVMeTCP/8d59b4b/DellEMC_PowerStore_NVMeTCP/386 : SUCCESS in 1h 46m 10s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"bbbb067fc6395d752a463cd06e1a60c2d9ad31b0","author":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"date":"2022-10-04 12:36:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"34f9474a69853dea3922b2ad755e56cc1b308b79","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-10-04 13:01:31.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6211b9089c695e3986e0be1bac0cd0a25f1f1955","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-04 14:14:25.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/47/859447/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5b50889/ : SUCCESS in 1h 36m 57s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"548ef095f5d318a98e2bdb7870d43a6589ee92ec","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-10-04 14:58:54.000000000","message":"Patch Set 2:\n\nFAILED \n\nhttp://openstack-ci.spb.yadro.com:8080/job/gerrit-trigger/735/ : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"22cd232a15fb3675fd56786a081a4b8b672527ac","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-10-04 15:38:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/47/859447/2/20221004210613/NEC-FC : SUCCESS in 3h 30m 19s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/47/859447/2/20221004210613/NEC-ISCSI : SUCCESS in 3h 32m 38s","accounts_in_message":[],"_revision_number":2},{"id":"656c0debea5fe969262a2d5e8da3d2319ee6dbac","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-10-04 16:38:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7108ef2a618ec2f3d7f96a0bce6ddfd51ed02e1e","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-10-04 16:41:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d0f8bd4e62410198dcbe990bd4141a43e40c4b6e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-04 17:27:27.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/47/859447/2/check/DellEMC_Unity_FC/ab4e9bd/DellEMC_Unity_FC/1896 : SUCCESS in 1h 09m 07s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/47/859447/2/check/DellEMC_Unity_iSCSI/f31b657/DellEMC_Unity_iSCSI/1743 : SUCCESS in 2h 09m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"ac77835069dcf830b875353ef5362a42d7abeac8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-04 18:35:32.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/47/859447/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a957c6f/ : SUCCESS in 1h 29m 24s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/47/859447/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/e10db49/ : SUCCESS in 1h 37m 17s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/47/859447/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/2394219/ : FAILURE in 1h 55m 54s","accounts_in_message":[],"_revision_number":2},{"id":"e35bf9dd61c3bd7e6f9fc6af1adaf02f51c9de01","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-10-04 21:38:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f28b576ae3e295211435968300aa1e410d1a079a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-04 23:57:11.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/47/859447/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/4b6a6b2/ : SUCCESS in 1h 28m 42s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/47/859447/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7156098/ : SUCCESS in 1h 39m 53s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/47/859447/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/25e42dc/ : FAILURE in 2h 18m 15s","accounts_in_message":[],"_revision_number":2},{"id":"d8a78b87acaabfca1f8653c5abe55e17bc65894f","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-10-05 13:26:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f5a5d44f64b0f1c05c969d6105b005622ee4991f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-05 15:08:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/47/859447/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/13cfd59/ : SUCCESS in 1h 39m 07s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/47/859447/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/b9bb12c/ : SUCCESS in 1h 40m 33s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/47/859447/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/b513462/ : SUCCESS in 1h 37m 24s","accounts_in_message":[],"_revision_number":2},{"id":"b7fc680c520aafb493a5bfef1c822153d9c67d11","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2022-10-06 09:16:07.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aa2980ad32a46d2ffab8f66582936d5e785fc22f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-10-06 22:22:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a253fa5b75185ddc712c51d73a501ae2aca66294","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 22:23:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"50316b7771eaf2dfc1bc25ca8ca87dda9e4ac479","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 23:49:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a18a1eeba974f3d9ef9c47e36164b77\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c012b8256461431f9a4b5ebc07eb8768 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89a49213a50f422a918fa5e70f3dc8ab : SUCCESS in 10m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07e38ceee4f64934b36ef992011cf68b : SUCCESS in 10m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/210c9470fcc645a597e32c681f194e37 : SUCCESS in 10m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/7bdbe319595c4b8dbc3f019dfbfcc3d3 : SUCCESS in 38m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2685aabbbe5246029468878a54c440da : SUCCESS in 1h 12m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7befb1db6a7f4ed3a838ef0ae80a774d : SUCCESS in 37m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e211893c8b1c4536bdcf126d498e6056 : SUCCESS in 59m 19s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9310a91d4146404e87315c6ac07bd6b1 : SUCCESS in 54m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5ab2c7eadc949899c4175fb6ce5fb66 : SUCCESS in 40m 47s","accounts_in_message":[],"_revision_number":2},{"id":"87d0bfbc11d39828b76303fd5d4beacb437867b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 23:49:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"06cade9bc41eef21a57d64215d16081ea2b87b1d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 23:50:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b019a497e8d14b579987f5f91d166f64\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a6b159b8ffb4f50930eda4ca3d3279a : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fae5538f919c420b682d8166547271c81ef4a3dd","revisions":{"99f116d37749520ec1ebae18560ba5bc4d49f486":{"kind":"REWORK","_number":1,"created":"2022-09-27 15:30:36.000000000","uploader":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"ref":"refs/changes/47/859447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/859447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/859447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/859447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/859447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/859447/1"}}},"commit":{"parents":[{"commit":"7b07379de1d80522fc814c8a14797deb13afdeab","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b07379de1d80522fc814c8a14797deb13afdeab"}]}],"author":{"name":"Keerthivasan","email":"ksuresh@purestorage.com","date":"2022-09-27 13:53:51.000000000","tz":-240},"committer":{"name":"Keerthivasan","email":"ksuresh@purestorage.com","date":"2022-09-27 13:53:51.000000000","tz":-240},"subject":"Pure Storage - bump version numbers for Antelope","message":"Pure Storage - bump version numbers for Antelope\n\nPure uses the version number of the driver to identify which\ncustomers are using which versions of OpenStack.\n\nThis will help granular analysis of which driver is being\nused in each OpenStack version\n\nChange-Id: I5b58e9485654f707740175d5ff2a60594dff42f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/99f116d37749520ec1ebae18560ba5bc4d49f486"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/99f116d37749520ec1ebae18560ba5bc4d49f486"}]},"branch":"refs/heads/master"},"fae5538f919c420b682d8166547271c81ef4a3dd":{"kind":"REWORK","_number":2,"created":"2022-10-04 04:37:47.000000000","uploader":{"_account_id":35316,"name":"Keerthivasan S","email":"ksuresh@purestorage.com","username":"keerthivasan"},"ref":"refs/changes/47/859447/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/47/859447/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/859447/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/859447/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/47/859447/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/47/859447/2"}}},"commit":{"parents":[{"commit":"7b07379de1d80522fc814c8a14797deb13afdeab","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7b07379de1d80522fc814c8a14797deb13afdeab"}]}],"author":{"name":"Keerthivasan","email":"ksuresh@purestorage.com","date":"2022-09-27 13:53:51.000000000","tz":-240},"committer":{"name":"Keerthivasan","email":"ksuresh@purestorage.com","date":"2022-10-04 04:37:07.000000000","tz":-240},"subject":"Pure Storage - bump version numbers for Antelope","message":"Pure Storage - bump version numbers for Antelope\n\nPure uses the version number of the driver to identify which\ncustomers are using which versions of OpenStack.\n\nThis will help granular analysis of which driver is being\nused in each OpenStack version\n\nChange-Id: I5b58e9485654f707740175d5ff2a60594dff42f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fae5538f919c420b682d8166547271c81ef4a3dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fae5538f919c420b682d8166547271c81ef4a3dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
