)]}'
{"id":"openstack%2Fcinder~860209","triplet_id":"openstack%2Fcinder~master~I1c1600edb7c1323e8ad7396b3e5dc4652b04f9f2","project":"openstack/cinder","branch":"master","topic":"860209","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-03-03 20:47:44.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2023-03-03 20:47:44.000000000","reason":"Change was submitted"},"30615":{"account":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"last_update":"2023-03-03 20:47:44.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-03-03 20:47:44.000000000","reason":"Change was submitted"},"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2023-03-03 20:47:44.000000000","reason":"Change was submitted"},"7198":{"account":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"last_update":"2023-01-05 15:04:08.000000000","reason":"\u003cGERRIT_ACCOUNT_7198\u003e replied on the change","reason_account":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-03-02 09:08:41.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I1c1600edb7c1323e8ad7396b3e5dc4652b04f9f2","subject":"Update IBM Storwize drivers","status":"MERGED","created":"2022-10-04 07:52:44.000000000","updated":"2023-03-03 20:48:58.000000000","submitted":"2023-03-03 20:47:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860209-860209","meta_rev_id":"30fb8da68ad75144f9cad7d7101e3233ba516de6","_number":860209,"virtual_id_number":860209,"owner":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-03 20:47:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"recommended":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"disliked":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":1,"date":"2023-02-22 05:25:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2023-02-22 05:25:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":-1,"date":"2023-02-22 05:25:02.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":1,"date":"2023-02-22 05:25:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2023-02-22 05:25:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-22 05:25:02.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":2,"date":"2023-03-03 17:41:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":1,"date":"2023-03-03 17:41:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32465,"name":"IBM Storage CI","email":"hchintha@in.ibm.com","username":"ibm-svf-user"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-04 08:22:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-04 08:52:19.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-04 08:56:06.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2022-10-04 08:56:57.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-10-04 08:58:00.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-10-04 09:02:29.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-10-04 09:28:30.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-10-04 09:35:37.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-04 09:47:00.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-10-04 09:59:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-04 10:33:30.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-10-04 11:24:11.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-04 13:11:29.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-10-05 02:15:25.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-17 09:43:09.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-17 11:11:28.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-24 06:46:43.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-11-02 05:24:39.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-11-02 05:28:06.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-11-02 05:28:33.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-11-02 14:05:45.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2022-11-02 14:05:45.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2022-11-02 14:05:45.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-11-04 14:39:21.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2022-11-10 16:47:53.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2022-11-18 14:24:24.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2022-11-28 11:52:16.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-11-30 11:22:52.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2022-12-16 16:28:27.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"REVIEWER"},{"updated":"2022-12-19 05:24:26.000000000","updated_by":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"reviewer":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"state":"REVIEWER"},{"updated":"2023-02-22 06:45:13.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-28 12:26:38.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-28 12:34:26.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2023-02-28 13:56:47.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-02-28 14:22:05.000000000","updated_by":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"reviewer":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"state":"CC"},{"updated":"2023-02-28 14:49:27.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-02-28 15:20:47.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-02-28 15:23:03.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2023-02-28 20:44:03.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-01 17:21:15.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-03-03 12:59:38.000000000","updated_by":{"_account_id":32465,"name":"IBM Storage CI","email":"hchintha@in.ibm.com","username":"ibm-svf-user"},"reviewer":{"_account_id":32465,"name":"IBM Storage CI","email":"hchintha@in.ibm.com","username":"ibm-svf-user"},"state":"CC"}],"messages":[{"id":"89249ed91d10875289dd571dbe479ebec5f6dfa8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-04 07:52:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4362e2815dbd2eddff9cfb5ea6fef890824410e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 08:22:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e4f69bd117247b9b314a59421c55552\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/31499bb8d79b462da4fc460d3372f3ea : FAILURE in 25m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/777c36a2cd9e4b87bab02da072f1c917 : FAILURE in 25m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e521b4abda255c0e09121153cd16e5dcc032ade2","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-10-04 08:52:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/09/860209/1/ : FAILURE in 51m 38s","accounts_in_message":[],"_revision_number":1},{"id":"af74b8049eb61d5a428d4583b03c1066949e0cd2","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-10-04 08:56:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/37869a1504a9474fad9946fafacaf481 : SUCCESS in 1h 02m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d42482255294cfc26f301196156808b7bb5683db","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-10-04 08:56:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/09/860209/1/check/iscsi-driver-alletra-6k/3ecc9fd/3ecc9fd.html : FAILURE in 1h 02m 59s","accounts_in_message":[],"_revision_number":1},{"id":"6fdd5857d7de9baf9af9caddcff45c2bec7353a5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-10-04 08:58:00.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/860209/1/iscsi/ : FAILURE in 1h 04m 45s\n- cinder-smb http://cloudbase-ci.com/cinder/860209/1/smb/ : SUCCESS in 53m 02s","accounts_in_message":[],"_revision_number":1},{"id":"342a26606ffa0a33bc174850b17b2694a688b2f8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-10-04 09:02:29.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/09/860209/1/check/dsvm-tempest-cisco-zonemanager-job/6e17782/ : FAILURE in 59m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4d21da2b886ce4da047dc9b4679a9b183dd2c9d0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-04 09:28:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/09/860209/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/cfaa928/ : SUCCESS in 1h 31m 15s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/09/860209/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7de7933/ : SUCCESS in 1h 27m 47s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/09/860209/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/813b783/ : SUCCESS in 1h 34m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e8045d8b610ca920f99eaaffa9f6ffbb7d0c9857","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-04 09:35:37.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/860209 : SUCCESS in 1h 41m 20s","accounts_in_message":[],"_revision_number":1},{"id":"1662dd154eff9940363e343d8d45d2712740657c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-10-04 09:47:00.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1664869972861 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"42c17bc05110c82eb0bc84e441846214d6bfb382","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 09:59:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9812eb2c71740ba85a5494c5fdb7d14\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2983a5d386b4167a3db0f7f1adfd807 : SUCCESS in 10m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8a5a9a575884d9dbe5ee0b3e0a40fcb : FAILURE in 11m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b416ea7fc41477a991bfd71313082c0 : FAILURE in 10m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fb86549cf984cca8912e8ebd6dde964 : SUCCESS in 16m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/8550d47ef77f486da89ae50812cd6c40 : SUCCESS in 59m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b6c19ccce30748b79d9d9ed1e91ecfc7 : SUCCESS in 1h 07m 03s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/09662abffa6549b282dcc5832d71db0f : SUCCESS in 1h 24m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b8f64ba880a42b694627471109fba13 : SUCCESS in 57m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0261456d717342f3911faad716365372 : FAILURE in 13m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ebeec05a7f664fa79b454556b00ba482 : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/60363ea8e1304d37bfc4c2addadacda0 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6bf7dbf9987b4a588f2172ca1ce03be4 : SUCCESS in 5m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f75d001c0e1e4510b9e029e7b7bb7212 : SUCCESS in 6m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a566ed11179d4cef96dbb7d6f33dab54 : SUCCESS in 45m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/47f8c3e1566947eea3db31c674ffb812 : FAILURE in 6m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6eef6efd5dd148f8a959ba5a814fba7c : SUCCESS in 1h 17m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1aa477409c6d42438f8ca6c2a984e8f6 : FAILURE in 1h 29m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c374e70be8ad40ffba3b37d6e0ec2559 : SUCCESS in 1h 51m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9d0b9158125f4ae7a6ffdda130d2d752 : SUCCESS in 1h 13m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7d030634d47b4cdda54935dba534d66f : SUCCESS in 1h 24m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d06af5eb35c342eaaf5692ada0165cfa : SUCCESS in 42m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7af6f17572a04d1dafc35119c43bcb63 : SUCCESS in 1h 31m 57s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/62ee4dec43e74c57b2fd3a6375656a7a : SUCCESS in 2h 05m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de3568eee62e46ccab973beb5be1381a : SUCCESS in 50m 17s","accounts_in_message":[],"_revision_number":1},{"id":"881860d7cd50fc17fd0387ba7aebc2a20403d038","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-04 10:33:30.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/09/860209/1/check/DellEMC_Unity_FC/d8b5055/DellEMC_Unity_FC/1892 : SUCCESS in 1h 11m 09s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/09/860209/1/check/DellEMC_Unity_iSCSI/a05ff7b/DellEMC_Unity_iSCSI/1739 : SUCCESS in 1h 56m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"9de7e607959c551c2115d88e3147806373d0e352","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-04 11:05:30.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"fcaea03f32066de8c8e06d4a673b3fbac6ac605e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-10-04 11:24:11.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-09-860209-1/ns5_nfs : SUCCESS in 3h 30m 57s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-09-860209-1/ns5_iscsi : SUCCESS in 3h 30m 56s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"0b3885b20fdd9697ce1141f36d2a45d802418745","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-10-04 13:11:29.000000000","message":"Patch Set 1:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/734","accounts_in_message":[],"_revision_number":1},{"id":"6d786d02c543c3b09c1953b6c857d761c233df90","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-10-05 02:15:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/09/860209/1/20221005041616/NEC-FC : SUCCESS in 3h 31m 23s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/09/860209/1/20221005041616/NEC-ISCSI : SUCCESS in 6h 59m 05s","accounts_in_message":[],"_revision_number":1},{"id":"13b8a9b32585548823db179fa54bbd4ef8091093","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-06 10:24:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d20921a7352be9cf4ebfbdc2f8a67251cd463b93","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-10-06 11:24:08.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/7156f707515040bd81af4f146ba15b9d : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":2},{"id":"20edf8ef932a1334a15646e35cb3d42f1e032a27","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-10-06 11:24:34.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/860209/2/iscsi/ : SUCCESS in 59m 31s\n- cinder-smb http://cloudbase-ci.com/cinder/860209/2/smb/ : SUCCESS in 56m 51s","accounts_in_message":[],"_revision_number":2},{"id":"d544612dd21f9f882e6e232386c228c3f8c3311a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-10-06 11:29:30.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/09/860209/2/ : FAILURE in 56m 42s","accounts_in_message":[],"_revision_number":2},{"id":"209d5f6cbf402dff63ead50150460e9dd6972567","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-10-06 11:31:42.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/09/860209/2/check/iscsi-driver-alletra-6k/5888b6f/5888b6f.html : FAILURE in 1h 03m 36s","accounts_in_message":[],"_revision_number":2},{"id":"0a22c0b133fcf277ced19f3d75639ef71a56d094","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 11:40:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c8d2ea0bcfc42aea1d03a17bfb01654\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db508775afb640e8bf4ac321006388e0 : FAILURE in 27m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2c42914e134f4716a78038d8ef05eb1f : FAILURE in 27m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cf7802ca222d1980abddc2d37b72e508139fe912","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-10-06 11:47:16.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/09/860209/2/check/dsvm-tempest-cisco-zonemanager-job/88781e8/ : FAILURE in 1h 13m 50s","accounts_in_message":[],"_revision_number":2},{"id":"1f9847fe5f07f002af83a684340edf0f6bf3166c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-06 12:01:38.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/09/860209/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/41efaf0/ : SUCCESS in 1h 36m 04s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/09/860209/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/558330a/ : SUCCESS in 1h 30m 18s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/09/860209/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/f94e974/ : SUCCESS in 1h 31m 58s","accounts_in_message":[],"_revision_number":2},{"id":"3b6efe7d236b683b66adf6eb3d518e951e6a26fc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-06 12:14:36.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/09/860209/2/check/DellEMC_Unity_FC/2ae4577/DellEMC_Unity_FC/1907 : FAILURE in 1h 12m 26s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/09/860209/2/check/DellEMC_Unity_iSCSI/f832d53/DellEMC_Unity_iSCSI/1754 : SUCCESS in 1h 49m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"4c1b40fba371d46748a1ba7479e0b2c6e650066a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-10-06 12:27:20.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1665051875730 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"d358aafb4888fbe7ab1d2dac08bc69ce771c9e51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 12:45:31.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/898ef7556c724cb78ce7be6b0eb1a48f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/232b6789b69946359e3dc8b05524925b : FAILURE in 7m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3e9a2b907e745cf8a2853d760894ec9 : FAILURE in 11m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/193d5e07348c46f8b79cf149bff30713 : FAILURE in 10m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68cb56610c204aeba84daf2287d06bd2 : SUCCESS in 13m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/2dd37b465b0947ac907d377cca02ae87 : SUCCESS in 1h 02m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ef1f721245894c718ac174540899e442 : SUCCESS in 1h 11m 57s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/361b6faaa3c8437392a4818971160b2c : POST_FAILURE in 1h 10m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55664b5835444f13b0902e31957eb3a2 : SUCCESS in 57m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6d50180278354d26b6bd8cc7a4a80afe : FAILURE in 15m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/51d5e9906dd84c77b59f20bf501a5bd8 : SUCCESS in 7m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2b1b0bd7f7d34a1ea847685cc127fd04 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1c6f63166b714c02b134c02027c93e26 : SUCCESS in 6m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/96effdf926484cd3a87cebd7a5394f89 : SUCCESS in 5m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/fac6a0e4e54a457e9462978d5cb9c85b : SUCCESS in 47m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/23984c4bae464c10b449df7b5fca813a : FAILURE in 5m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/afdd1cb55da34de0badad49704fba210 : TIMED_OUT in 2h 11m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9d555a450c014af58ca1418538677543 : FAILURE in 1h 37m 04s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ea534ee3cb52407d821d0a4f37bf896c : SUCCESS in 1h 47m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e29bd53aa033445aba1e243f11b3c308 : SUCCESS in 1h 12m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2df683a5e8fa4523a06faf510307842f : SUCCESS in 1h 31m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/beb923132f7d42579dea44a20051b8e1 : SUCCESS in 47m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e36f6afac39f487e803356a2bbfa96b7 : SUCCESS in 1h 30m 46s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/abc6d900bad643a9b4e112409af2e4be : SUCCESS in 2h 14m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c049a8d872cd4f5182db675be48698cf : SUCCESS in 54m 56s","accounts_in_message":[],"_revision_number":2},{"id":"cc1b6d76fd771df114a84d6a86c887a6ef5ecf82","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-06 12:51:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/860209 : FAILURE in 2h 25m 16s","accounts_in_message":[],"_revision_number":2},{"id":"874e5a436e63e2aae84734245c7a700ebd5ac2a4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-10-06 13:57:38.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-09-860209-2/ns5_nfs : FAILURE in 3h 30m 2s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-09-860209-2/ns5_iscsi : SUCCESS in 3h 32m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"8347255939849395a09da18bf7df873b322b2c1c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-10-06 20:35:12.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/09/860209/2/20221007020044/NEC-FC : SUCCESS in 3h 29m 45s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/09/860209/2/20221007020044/NEC-ISCSI : SUCCESS in 3h 34m 25s","accounts_in_message":[],"_revision_number":2},{"id":"7e6c33066b3a0b9f090fded3e1d146397da69faa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-10 06:03:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5cd8a88781411e5e8441f7f1dc78b619c2da1af2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 06:34:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6139ef9543a45f988339c5497799ecd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/41d0850340e54a268cf35cfcec6a89b9 : FAILURE in 25m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2166c68de5674b30b51291b801bfc2ce : FAILURE in 26m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"800e7c232d5c10f73298e44b5a45cd6e21d4e296","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-10-10 07:03:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/3e7ee880214c46a98249fc5209efb1b5 : SUCCESS in 59m 22s","accounts_in_message":[],"_revision_number":3},{"id":"17a0ca0395dbb06f10c6e18166e8b85756c24087","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-10-10 07:08:49.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/09/860209/3/check/iscsi-driver-alletra-6k/9b1cbbd/9b1cbbd.html : FAILURE in 1h 03m 58s","accounts_in_message":[],"_revision_number":3},{"id":"68cdc7ee9d02f4ff0acb152cbde97560fbc86026","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-10 07:34:48.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/860209 : SUCCESS in 1h 29m 01s","accounts_in_message":[],"_revision_number":3},{"id":"db54a0174633f93df25424c6ba3a003e7ed0b140","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-10-10 07:41:02.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/09/860209/3/check/dsvm-tempest-cisco-zonemanager-job/fa4c84c/ : SUCCESS in 1h 17m 09s","accounts_in_message":[],"_revision_number":3},{"id":"48c372b69120bbc48bec405f072c0842f5420a72","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-10-10 07:42:38.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/09/860209/3/ : SUCCESS in 1h 30m 03s","accounts_in_message":[],"_revision_number":3},{"id":"0a2ae6f44e2a84a8c79de452bd886e1094ab8cb2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-10-10 07:48:03.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1665381845341 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"2cd793351f33495fb849686fd6441c95548c7121","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 07:56:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f655ea2108fc49a3ab6dede37498f8a3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7a3c5550d5b4065b9b3128860d237e6 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba0f6a33cdf24917a64ae7bbfdc740df : FAILURE in 11m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c48a6fbb9a15479aba8546c78b1ad798 : FAILURE in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48ffa8cc5e04435eba0dffa6f7576b9e : SUCCESS in 11m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/650e161cef644ec69d02c5f3d33b17ce : SUCCESS in 59m 32s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c77a4206b70c4ae586d008878b233489 : SUCCESS in 1h 02m 32s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4136829d56c94d6ea10d618dfa598a9d : SUCCESS in 1h 25m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e2bdf011a5243eda0db1fb6e969152a : SUCCESS in 55m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/135942841c5c4534b0c958a052b31e5e : FAILURE in 12m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f16b57ee76c54e11b462646f89c5e187 : SUCCESS in 4m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/42678bf3df8a4c10b76dc1ee337814db : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d50d6fbf90c6448d88ab16db88450a46 : SUCCESS in 5m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2adce0c8a0be49a8afc62b38be1d2be4 : SUCCESS in 6m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0e623e10f5e74474b3d6d1839ea7b057 : SUCCESS in 42m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/51b01c242fca4d179845a270c025e833 : FAILURE in 5m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d3c2b83d1d8c437faa2c1c78e010bc66 : SUCCESS in 1h 05m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/382b9b05a43d4f0985d1232c534fcf61 : FAILURE in 1h 12m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7fa3938ae11247ec8a0c520d89c7c78b : SUCCESS in 1h 24m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f5360a6f12594b75a63647dbe86a6150 : SUCCESS in 51m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5c6efb6f1ed84b46b2fada2429841851 : SUCCESS in 1h 04m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c49acae7bfe44a2989aa33bcae96877c : SUCCESS in 32m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/03bb7b70e18142b19716551dfec0aef1 : SUCCESS in 1h 16m 30s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1e8150879c8844b5a8402622cb011e19 : SUCCESS in 1h 45m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c049db41b6e94e4baf9737a9fe59f5c8 : SUCCESS in 39m 28s","accounts_in_message":[],"_revision_number":3},{"id":"da66555d4134d4e5e5a9be106ce967df64c220b3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-10 07:58:34.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/09/860209/3/check/DellEMC_Unity_FC/c3e5c9c/DellEMC_Unity_FC/1918 : SUCCESS in 1h 20m 17s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/09/860209/3/check/DellEMC_Unity_iSCSI/765c244/DellEMC_Unity_iSCSI/1765 : SUCCESS in 1h 54m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"d08a7ec2e700dea432114e4a35b61c1a2d7f57d3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-10 08:24:51.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/09/860209/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/284b296/ : SUCCESS in 1h 40m 26s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/09/860209/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a17c945/ : SUCCESS in 1h 32m 58s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/09/860209/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/b12ce94/ : FAILURE in 2h 19m 52s","accounts_in_message":[],"_revision_number":3},{"id":"3e9cddb2d510f1e2e8840d8e93a4aa8f4ae8f9df","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-10-10 09:31:35.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-09-860209-3/ns5_nfs : SUCCESS in 3h 27m 13s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-09-860209-3/ns5_iscsi : SUCCESS in 3h 23m 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"267fd95065683339b5ad708c6985102042152213","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-10-10 09:39:04.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/09/860209/3/20221010150356/NEC-FC : SUCCESS in 3h 33m 49s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/09/860209/3/20221010150356/NEC-ISCSI : SUCCESS in 3h 35m 04s","accounts_in_message":[],"_revision_number":3},{"id":"543e89fcbe58179701d2626e178539861bf991d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-10 11:11:12.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"a25c2a6a2f411e93153617c9664ded2fb04c0e4a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 11:35:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e1ba04e9fca4e6b83065982bb15d9b8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8dbff33c73c54fb796ab627ea3945402 : FAILURE in 23m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8b5d4bc026b94bfc988b14ba6cc148d3 : FAILURE in 13m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"791626d035f1bf54d46dc588df8f0ce0210e772c","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-10-10 12:11:51.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/b9d4d2b7f7d246959751234be6d878ab : SUCCESS in 59m 50s","accounts_in_message":[],"_revision_number":4},{"id":"8eb7b29ecde902507fb412f0cfb6be84772f5053","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-10-10 12:15:42.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/09/860209/4/check/iscsi-driver-alletra-6k/c28c46f/c28c46f.html : FAILURE in 1h 03m 33s","accounts_in_message":[],"_revision_number":4},{"id":"759dd386b035d207d51eae6e858647ead08d8cbd","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-10-10 12:26:04.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/09/860209/4/check/dsvm-tempest-cisco-zonemanager-job/baf8f53/ : FAILURE in 1h 01m 17s","accounts_in_message":[],"_revision_number":4},{"id":"7c509ee18416651dfd4a6f4aa0502a22b98a27be","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-10-10 12:28:06.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/860209/4/iscsi/ : FAILURE in 1h 15m 03s\n- cinder-smb http://cloudbase-ci.com/cinder/860209/4/smb/ : SUCCESS in 1h 16m 16s","accounts_in_message":[],"_revision_number":4},{"id":"f744a38eb1678bd69336dfb49dfd16b7f90f4a0e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-10-10 12:56:03.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1665400280892 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"9e04744e0c3254f6f7cc944f0949bf02c708212e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-10 13:05:26.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/09/860209/4/check/DellEMC_Unity_FC/5a11f49/DellEMC_Unity_FC/1919 : SUCCESS in 1h 11m 29s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/09/860209/4/check/DellEMC_Unity_iSCSI/afdbb59/DellEMC_Unity_iSCSI/1766 : SUCCESS in 1h 53m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"ef8daf4d450b4bcaae9e7339105c04b73644f6a2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-10 13:11:06.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/09/860209/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2c5821e/ : SUCCESS in 1h 43m 52s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/09/860209/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d034e44/ : SUCCESS in 1h 58m 57s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/09/860209/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/6d6924d/ : SUCCESS in 1h 35m 49s","accounts_in_message":[],"_revision_number":4},{"id":"20b9d735add54f20294f8bc4896cbfa141c46e1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 13:16:23.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a68a95de2bcc43bd8bbbd0b6f831fe26\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c619ee5909b2492f823385949f01345d : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/824781293f76429ea8c0671f92c75c4b : FAILURE in 8m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9cb7ad6e39c0495696aeb796ecaee1b3 : FAILURE in 8m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbc2843ade9c41de93e5dc3d0278e0f2 : SUCCESS in 10m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/10e83f44a4f940a184a63c005da8a068 : SUCCESS in 43m 21s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bb10bc02afce4981962c36d9ac8001fa : SUCCESS in 40m 51s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/826e0f80cac3479f819dc002a9bbe11b : SUCCESS in 1h 06m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b25d0487f8534154a78c0fe9f650d768 : SUCCESS in 56m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b065cb4f3fb545e6b95200b30ed18643 : FAILURE in 13m 18s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/246ffc9d21224c4fa239133dcc7df8ca : SUCCESS in 5m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6df1d26708f34f8587cbc6d43174fa73 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7f2af2b1b7d6484eb63bb7d9dfbe09be : SUCCESS in 6m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/848d3a8de05d41efa3315be082b36aa5 : SUCCESS in 7m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/967b3407c06a47828e87b7c2f7848f56 : SUCCESS in 43m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2c3f11d96fd04ec4aeb437bac89c0699 : FAILURE in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/43e9683693964de98c6d39bd4f7c10af : SUCCESS in 1h 11m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/845f64fa77234b00b55b0f69ed1f022b : FAILURE in 1h 15m 43s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4f8e3275ed984a388de568c0e53f0071 : SUCCESS in 1h 53m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/947efb562eb84525bfcbf979f8253f54 : SUCCESS in 1h 00m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d5abacf799af4faf91da062a414f5b77 : SUCCESS in 1h 28m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fe63e164cf2c4167b16822cdefb47e31 : SUCCESS in 44m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e5f6a2e8d1164294b8dfb77fad6cf7d5 : SUCCESS in 1h 27m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6f22413b63034a37966bc03d1cd7907a : SUCCESS in 2h 02m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32e5aed0a1eb4d988610b4d44d537ce8 : SUCCESS in 50m 32s","accounts_in_message":[],"_revision_number":4},{"id":"95c6c716ac544f828fdb4c1052cb2024e0c1d1dc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-10 13:27:06.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/860209 : FAILURE in 2h 14m 21s","accounts_in_message":[],"_revision_number":4},{"id":"ef38bba396cfcf78824271f844e11558a657bae4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-10-10 13:52:03.000000000","message":"Patch Set 4:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/746","accounts_in_message":[],"_revision_number":4},{"id":"b4eb96d65be7df65b2fa2429baea52bafd2e0cc6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-10-10 14:39:50.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-09-860209-4/ns5_nfs : SUCCESS in 3h 28m 9s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-09-860209-4/ns5_iscsi : SUCCESS in 3h 26m 38s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"15e3f46d5335a1e9462ff52c702efe8a34a7e8e0","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-10-10 14:44:20.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/09/860209/4/20221010201114/NEC-FC : SUCCESS in 3h 31m 51s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/09/860209/4/20221010201114/NEC-ISCSI : SUCCESS in 3h 33m 03s","accounts_in_message":[],"_revision_number":4},{"id":"a74091d887850137d2b4bc2f1c77360df8d5f693","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-10-10 14:59:05.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/09/860209/4/ : SUCCESS in 2h 07m 53s","accounts_in_message":[],"_revision_number":4},{"id":"34435d7ff6e12b85c157c5d19ab77ed15a50e4d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-17 09:42:27.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8d60403c957fd51564c85568808f1563740b38ec","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-10-17 09:43:09.000000000","message":"Patch Set 5:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/09/860209/5/check/fujitsu-eternus-dx-iscsi/9558e3b : NOT_REGISTERED\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/09/860209/5/check/fujitsu-eternus-dx-fc/34694c2 : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"0711ed51d4780dd3ea939ed00f490204486afbb0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 10:00:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20e6145492914c6bba8669ecd32495c2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ebcc2b247d934160b0c584d276495253 : FAILURE in 13m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c0e522bcfdd543c0afc1bcfcfe07465a : FAILURE in 13m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"17037147ec88e77aab138d9d034e389937e254ad","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-10-17 10:24:29.000000000","message":"Patch Set 5:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1665999753074 : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"a4038fdb5a89831d85f976c6d2b4970c806a27a2","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-10-17 10:40:20.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/7e535e4aa9fd43f6b2f5a46250dff83c : FAILURE in 57m 10s","accounts_in_message":[],"_revision_number":5},{"id":"d02d31f1fd0c23f8a721ec56bba833907a1bd2f9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-10-17 10:43:39.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/860209/5/iscsi/ : SUCCESS in 1h 00m 40s\n- cinder-smb http://cloudbase-ci.com/cinder/860209/5/smb/ : SUCCESS in 55m 47s","accounts_in_message":[],"_revision_number":5},{"id":"8aa14e3b9b28be1956cc352ca63655d781f8849f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-17 11:04:54.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/860209 : FAILURE in 1h 21m 06s","accounts_in_message":[],"_revision_number":5},{"id":"138e227e227360a39a1ad72027719ac794b196ca","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-10-17 11:08:54.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/09/860209/5/check/dsvm-tempest-cisco-zonemanager-job/1c6ea2a/ : FAILURE in 1h 03m 05s","accounts_in_message":[],"_revision_number":5},{"id":"86efe666947887918b31c75ab088a9e0e7ae855c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-17 11:11:28.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/fa1eb18bd4284a52b59efb33e76ab632 : POST_FAILURE in 1h 16m 30s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/374a7d8de83a4e568ac98af119554936 : POST_FAILURE in 1h 25m 49s","accounts_in_message":[],"_revision_number":5},{"id":"306b6e43e54b30b09486df145decbc9e0aa785f9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-17 11:26:38.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/09/860209/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b0d4c1c/ : SUCCESS in 1h 35m 56s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/09/860209/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/59cff84/ : SUCCESS in 1h 41m 03s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/09/860209/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/56ac5d7/ : SUCCESS in 1h 42m 50s","accounts_in_message":[],"_revision_number":5},{"id":"4a7da07f9492816061fa299d6cbd325bdf9a9d20","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-17 11:45:19.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/860209 : SUCCESS in 2h 01m 15s","accounts_in_message":[],"_revision_number":5},{"id":"6d6f8941255d50de0d862ceacea9befffea96cb9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-10-17 12:05:57.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/09/860209/5/check/iscsi-driver-nimble/a29bffb/a29bffb.html : FAILURE in 1h 38m 24s","accounts_in_message":[],"_revision_number":5},{"id":"a0f3ec6702feec9a3f2a847e2ff9ea3cf03fb855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 12:07:14.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6dbb35d928f41b280ec8eca70848b96\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e96ea9c355514f92833b27290194f5e6 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19b6c89e56734cbbbd66c534074098ef : FAILURE in 10m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ea8d636c554486ab564c3ddef24c897 : FAILURE in 9m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92fa25eb10e54e798f399794e0691be4 : SUCCESS in 13m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/816639ea243140bcad0e9cb64b1c4452 : SUCCESS in 54m 02s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b9858e7a7ca649a8bf071e7e2cb30520 : SUCCESS in 1h 03m 35s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/da141d7a3002407581d86baa994487c8 : SUCCESS in 1h 32m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1146a80626b49348c96bc686da058e7 : SUCCESS in 55m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1c85195010a24acfa9dad9ce2d037318 : FAILURE in 12m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ac0c9bdeb6734f43ab84a7560816cd4a : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/11a395112c7f40f29542942d50cc6e0e : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e9635348da7c49889cf17ddb7d8db980 : SUCCESS in 6m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c45091078c2e44ca8d30934a63c5f197 : SUCCESS in 5m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a285ebd4674e47b485b011fb860c259c : SUCCESS in 43m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c8529eeebcd74fde9a85b63527d43d2e : FAILURE in 5m 40s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2cbe33887f5148f29b4c8f5001bff3e9 : SUCCESS in 1h 15m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cb9f98b67d0042548f4563453bc0af4e : FAILURE in 1h 19m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/307be4e71faf47b1af3bf7683f148197 : SUCCESS in 2h 01m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e5b31bb5b8e4434a9f671532322eebf2 : SUCCESS in 1h 01m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b1c72f92d602406783ddf44b171f8d44 : SUCCESS in 1h 24m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/953de63d418a4474933345dc219fdcc7 : SUCCESS in 40m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9b5465dd254d429db21a6bf604da956d : SUCCESS in 1h 24m 37s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b9acbf95482344b69000c504076b99cc : SUCCESS in 2h 17m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3afb3cd2417b41f99776bfb21d64e9a0 : SUCCESS in 51m 31s","accounts_in_message":[],"_revision_number":5},{"id":"7d29a19192bef85a0f751c7f41015cb8f9e9200d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-10-17 13:14:18.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-09-860209-5/ns5_nfs : SUCCESS in 3h 28m 4s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-09-860209-5/ns5_iscsi : SUCCESS in 3h 31m 21s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"a412eeb8189fdf76814d5143118678bb527b3a8c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-17 14:52:29.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/09/860209/5/check/DellEMC_Unity_FC/295844c/DellEMC_Unity_FC/1951 : SUCCESS in 1h 07m 57s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/09/860209/5/check/DellEMC_Unity_iSCSI/71cb1f5/DellEMC_Unity_iSCSI/1798 : FAILURE in 2h 01m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"eaa0fe28bcd39879160769d17d9fc86927e55399","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-10-17 17:20:24.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/09/860209/5/check/iscsi-driver-alletra-6k/b399be3/b399be3.html : FAILURE in 1h 39m 31s","accounts_in_message":[],"_revision_number":5},{"id":"bb2e24e86ffe68876db5f6b2866e0b2e282810ee","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-10-18 10:28:12.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/09/860209/5/20221018141817/NEC-FC : SUCCESS in 3h 21m 26s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/09/860209/5/20221018141817/NEC-ISCSI : SUCCESS in 5h 09m 51s","accounts_in_message":[],"_revision_number":5},{"id":"9120f5b6bc2d3a9e5386a2fd89c7a4fa872235b3","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-10-20 14:10:34.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/09/860209/5/ : SUCCESS in 1h 40m 18s","accounts_in_message":[],"_revision_number":5},{"id":"49b6495e872a0a5cce37eab3d3f1508a98df4e67","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-10-24 06:46:43.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/09/860209/5/20221024094442/NEC-V-ISCSI : SUCCESS in 2h 42m 26s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/09/860209/5/20221024094442/NEC-V-FC : SUCCESS in 6h 01m 58s","accounts_in_message":[],"_revision_number":5},{"id":"95e9c67b5b66f970d2c0ec1e602afa0cd241452d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-11-01 16:40:53.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"52d4f559172badfbe96e05864a27ac1d120913f5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-01 16:41:27.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"c0ffc00565c985521444d6220eb1b436985c2627","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-01 17:09:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/083bae529ec04d5697543ffcec7a9cba\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/71dd45ea0df84b949724eb651a6e3b93 : SUCCESS in 26m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7e96677f7c5a484985b8f497ae988cef : SUCCESS in 20m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"db41207041860e2c457798705c462c3f59bda04f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-11-01 17:34:46.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/860209/6/iscsi/ : SUCCESS in 53m 22s\n- cinder-smb http://cloudbase-ci.com/cinder/860209/6/smb/ : SUCCESS in 50m 16s","accounts_in_message":[],"_revision_number":6},{"id":"e2afb1e09ee6fdea18c4263bd320b47f56976b01","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-01 17:49:58.000000000","message":"Patch Set 6:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1667320863332 : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"ebf7942ea79ccaa0caca5f432086590cc9456e2e","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-11-01 18:06:16.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/a51fbaa9ddba4f0dac107b2dca6c44ad : SUCCESS in 1h 24m 40s","accounts_in_message":[],"_revision_number":6},{"id":"d7198012117423043da69f45859ea903068823cd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-01 18:11:35.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/860209 : SUCCESS in 1h 27m 49s","accounts_in_message":[],"_revision_number":6},{"id":"afb96b88398d67f977cb97e7dcf6473b46daa252","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-01 18:14:58.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/860209 : SUCCESS in 1h 27m 55s","accounts_in_message":[],"_revision_number":6},{"id":"cf58d74307e4a3f7bf5d90c7b79c9294c358f030","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-11-01 18:19:43.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/62ff1e2147fe41ffbdf038e23c547722 : POST_FAILURE in 1h 16m 38s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/fef8e71e4be543988ba85d9e7a051786 : POST_FAILURE in 1h 35m 43s","accounts_in_message":[],"_revision_number":6},{"id":"365ae3a23b53d3e966f27c7f6a73eaa14d0c387e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-01 18:37:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/50f8ae48b8bb472a92c1b8e11778f655\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/252711b4d6284723835f1de36df65f73 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97509adc0874472fa082f0e6cf6400af : SUCCESS in 8m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df8e11b959944d7995582a65f99bccfb : SUCCESS in 9m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba20d9deb80b4983a3feb9316393da3a : SUCCESS in 12m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/08a3317f93fd4bc0b2587d8307511cb3 : SUCCESS in 35m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ca8bb78463e540df922632f52429ad3d : SUCCESS in 51m 41s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c6c80f12d9384d408b31fb89bb5c25aa : SUCCESS in 1h 16m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/147ce427a0fc4e5aa2379627ba7687b6 : SUCCESS in 38m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4c714d4191e24628a0469e8d5f3f5f9a : SUCCESS in 15m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ccbd6901d5cc42b7961a93b0ba1eb2b7 : SUCCESS in 5m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cc883bf5a44a4093a330b8790a85da1d : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7d84cf7ae72843ffb88f038fda92b1c5 : SUCCESS in 5m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6abf3ddc5b2b460791f7e9ed4467f3de : SUCCESS in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/27eeb7ceb70846f9a2e4f9956e48a569 : SUCCESS in 41m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/49235f709a9d4deca7fad036535424ae : FAILURE in 6m 09s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b17ae3ba85784123b3f8e3e3dad666c3 : FAILURE in 1h 00m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3084b1ce04ee45649867c578acfda9ee : FAILURE in 1h 28m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c56b29aa6aaa44c8a0a3fdd4eb6ffee3 : SUCCESS in 1h 50m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c00c84cb3fc9467a868a0d5042d166fc : SUCCESS in 48m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/335fe92061d24ce19bf28e224c174451 : SUCCESS in 1h 07m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f8df6de3b2b647b298a060e3d095586b : SUCCESS in 39m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8f55b279662548ad901066651927150d : SUCCESS in 1h 38m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/180f4b1ac71a4a07b54135012f183f98 : SUCCESS in 1h 30m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b509257f1bdb45e0a7536f7485c55172 : SUCCESS in 47m 18s","accounts_in_message":[],"_revision_number":6},{"id":"c1aa381bf173ed8587dba9411692a48b31759277","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-01 18:48:17.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/09/860209/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/35df7c5/ : SUCCESS in 2h 01m 49s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/09/860209/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/9ff8038/ : SUCCESS in 1h 37m 50s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/09/860209/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/773d155/ : FAILURE in 2h 06m 12s","accounts_in_message":[],"_revision_number":6},{"id":"257a75c2226d86177ae1166cf4eaed14f547426d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-01 19:10:07.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/09/860209/6/check/iscsi-driver-alletra-6k/812164d/812164d.html : FAILURE in 2h 23m 10s","accounts_in_message":[],"_revision_number":6},{"id":"cd18fb7c96d116bcc0020c1c1fc123fedc88a274","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-01 20:10:02.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-09-860209-6/ns5_nfs : SUCCESS in 3h 28m 39s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-09-860209-6/ns5_iscsi : SUCCESS in 3h 28m 38s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"f5b8f8db1aeb290bf0afcb7be7262050b553f4ff","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-11-01 20:17:32.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/09/860209/6/20221102014053/NEC-FC : SUCCESS in 3h 35m 31s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/09/860209/6/20221102014053/NEC-ISCSI : SUCCESS in 3h 36m 35s","accounts_in_message":[],"_revision_number":6},{"id":"a8877030216b90c038253edf800d006bac057a02","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-11-01 22:13:41.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/09/860209/6/20221102014056/NEC-V-ISCSI : SUCCESS in 2h 27m 05s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/09/860209/6/20221102014056/NEC-V-FC : SUCCESS in 5h 32m 42s","accounts_in_message":[],"_revision_number":6},{"id":"82472f888e0534c9e81e10258b483985f0787b22","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-11-02 02:52:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b8ef8d7d0bf64baa1bc19399696e222eaf9acfc0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-02 02:53:57.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"49320c36f3684847647494a0a6c84ed6dfb97cf1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 03:55:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaadaba02789440c93b9b6d0beed832d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8b131d4d883647caa0564a962e814a6b : FAILURE in 27m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/675cb11163474f20aca3da565e6055cf : SUCCESS in 20m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3a729a5cb78139cef0bfa67e24b664cfe24be60c","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-11-02 04:15:02.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/10beae1849ac45f79a346a3a13534192 : SUCCESS in 1h 20m 34s","accounts_in_message":[],"_revision_number":6},{"id":"62144d5bfc38c241b315def04363cb385724d1ff","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-11-02 04:19:42.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/860209/6/iscsi/ : FAILURE in 1h 26m 26s\n- cinder-smb http://cloudbase-ci.com/cinder/860209/6/smb/ : SUCCESS in 56m 08s","accounts_in_message":[],"_revision_number":6},{"id":"3af960e5d39d5dbad99671a85bef56ea8e5996bb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-11-02 04:29:51.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/f2a3d7345e40455bbd7134bded602f5b : POST_FAILURE in 1h 20m 40s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/1e71e6b004094438b5ee862321c57903 : POST_FAILURE in 1h 34m 09s","accounts_in_message":[],"_revision_number":6},{"id":"fdcecf4629c5088093f868f406f713a829698ce8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-02 05:04:44.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/09/860209/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/e6a49d1/ : SUCCESS in 1h 46m 13s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/09/860209/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a023198/ : SUCCESS in 1h 47m 41s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/09/860209/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/bb65ecb/ : FAILURE in 2h 10m 36s","accounts_in_message":[],"_revision_number":6},{"id":"d3cfc23f0a3389c23524af2856c4c25681ff1635","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 05:16:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc24c08bdb81461382042d6b80fa9615\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b084ab71c614eb494f1ac8352c3b150 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18a9ccce032a42d697b89c199a01da1b : SUCCESS in 10m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c8aa3307fd04d619aa66347dd59fc2b : SUCCESS in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ca0e18d18934f189bc75a45a399e1f8 : SUCCESS in 10m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/f949cf5714ac42b4b509fbbdbebf5972 : SUCCESS in 59m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5d20895b0cb84ed98048623fa49131df : SUCCESS in 1h 05m 58s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f26f14e854eb49d3bdaa6ba18777d815 : SUCCESS in 1h 25m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35e6435808e84240806dc09414aee9ae : SUCCESS in 48m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cbd78d5660104e10b9fa9e88dc47ef80 : SUCCESS in 17m 15s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a00f132f9c3f4a7e98b56c6fd7549c23 : SUCCESS in 5m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8238cd54a2834bcd8c52729ee2d33e83 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7559c56af2204c388b6b36cbc0ad570f : SUCCESS in 6m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0ec22c963ced4c909e07494d77a36cab : SUCCESS in 6m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8a0b9d1e41bf46c38c66e13a085a18a0 : SUCCESS in 33m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cde6a4c87f054cc490a2b70b7709a4f6 : FAILURE in 6m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dbbf200f31514d9d9ab8ef3b748337da : SUCCESS in 1h 06m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f26dbd97879d4b79ae37affd3b0c6a86 : FAILURE in 1h 40m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9d271921e9d0463eac0cb34cd13271d9 : SUCCESS in 1h 27m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5a726f4a4e604d3cb90a69968f9a9253 : SUCCESS in 57m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/618e11c24b6f4acb81729547870b473f : SUCCESS in 1h 14m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/71983fc3583f417f9843f0d495e8fa00 : SUCCESS in 36m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d64fc6f5bd55417c900e0070078421e1 : SUCCESS in 1h 20m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e20eadff05844936ac0da3c964c52ecb : SUCCESS in 2h 18m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64672d3c00a9483a9d15d69a31142ca3 : SUCCESS in 57m 37s","accounts_in_message":[],"_revision_number":6},{"id":"9f1cfcf14ef90797e4827ef17ce57d7c0ba6db4f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-11-02 05:24:39.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/02_11_2022_05_17_00\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"1dfdf12f1a26e3c63a839ed55fafa5a92ed12714","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-11-02 05:28:06.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/09/860209/6/upstream-check/cinder-tempest-plugin-ontap-fcp/068f8e7/ : RETRY_LIMIT in 9m 06s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/09/860209/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/724bcec/ : RETRY_LIMIT in 9m 11s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/c74fbecaa5f9455597eca8a7d7270ae1 : RETRY_LIMIT in 2m 15s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"fcb71c2e785578237f8f108744c27fc613b8fdbd","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-11-02 05:28:33.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/09/860209/6/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/bb09ef3/ : RETRY_LIMIT in 8m 47s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":6},{"id":"33805ce7b8ae0d5da799243a871857ee6955c670","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-11-04 14:39:21.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"54036a621a6b384625cb695f838b4b8b6f872d46","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-11-05 22:35:31.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/09/860209/6/ : FAILURE in 1h 38m 13s","accounts_in_message":[],"_revision_number":6},{"id":"1c2f13c48e8a97eeb7f570964e749c768da1721a","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2022-11-10 16:47:53.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"18ca31a88b4aeb4c4416c66dfe4a66d448e9fbeb","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-11-18 14:24:24.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"70ca8cffc875778a3f09fd4cfcb39c9f1f111fa4","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-11-18 14:24:57.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2b8383bdbc1ad67369add12a204725b0bcafa271","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-11-28 11:52:16.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"fb3338a881ad2fe5854a0b86ad20509b62555750","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2022-12-16 16:28:27.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c31c26a764fc016b9d56b48bd452d594e024b9bf","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2022-12-19 05:24:26.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5145b456789df81f75d9f2a4ccdcf922b67fbc87","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-12-19 19:35:07.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fd8fa02191f1d8455034dc91030466c41a9ab547","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-12-19 21:35:56.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/09/860209/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/8da9aaa/ : SUCCESS in 1h 34m 53s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/09/860209/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d892894/ : SUCCESS in 1h 57m 03s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/09/860209/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/a969e79/ : SUCCESS in 1h 59m 36s","accounts_in_message":[],"_revision_number":6},{"id":"2ef9b4aba0b09a0089e4158ad69939b2367a39c1","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2023-01-05 15:04:08.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0986b3d187ae99142bbbf0a6d53e42e0aa3a13ed","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2023-01-19 16:42:46.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0bb53ca0deb4aaf8972c660c0468b636bf3706d8","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-02-15 14:15:19.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"37329e15418ae792087c0ae37a3d49cd14c4ba70","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-02-22 03:28:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9ae6c9902b0e8313db1727e717c3816b5bf1c5de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-02-22 05:25:02.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"c69d9fc1dd71c00432ab70d2f1d47d842e22cfc0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 05:48:47.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f70e9e0a3ae485096fa166da6564dbb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d1d6812cd704a34af5dcad87d0b739f : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9c2eae9567d14efb9da45cbfe10c358a : SUCCESS in 21m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e6e1706a9fe782d99208c06abe89d1c68518843d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-22 05:49:53.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/860209 : FAILURE in 23m 29s","accounts_in_message":[],"_revision_number":7},{"id":"ae44cb3a6f703c94da105ac33f694f49a38b30f8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-22 05:52:02.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/09/860209/7/check/iscsi-driver-alletra-6k/6df02ba/6df02ba.html : FAILURE in 25m 37s","accounts_in_message":[],"_revision_number":7},{"id":"1b52c540bfc73ace23b95fcc6f83f0c4b7e8d4e9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-22 05:52:45.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/860209 : FAILURE in 25m 59s","accounts_in_message":[],"_revision_number":7},{"id":"78f33d8c22d2f46211722de82cb4ae917033e3d6","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-22 06:29:20.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/09/860209/7/ : FAILURE in 56m 20s","accounts_in_message":[],"_revision_number":7},{"id":"f308144bd8aa1980f57874c830271e9fd26c86e6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-22 06:29:56.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-09-860209-7/ns5_nfs : FAILURE in 1h 3m 49s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-09-860209-7/ns5_iscsi : FAILURE in 1h 4m 26s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"5273146d8fbefd4643fddc99900992353ed40c35","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-22 06:45:13.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/50f36180adfa4e2bb1ecc65c5b76fae6 : SUCCESS in 1h 18m 13s","accounts_in_message":[],"_revision_number":7},{"id":"4434e8e576f583ebf39d079848c7ba550224c672","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-22 06:46:04.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/d9c693b61660467ea0b075963af32cc3 : SUCCESS in 1h 20m 23s","accounts_in_message":[],"_revision_number":7},{"id":"2e9b5b007a0df02065b0d26247556c625fea043a","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-02-22 06:55:57.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/09/860209/7/check/fujitsu-eternus-dx-iscsi/cb7b9c4 : SUCCESS in 1h 30m 17s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/09/860209/7/check/fujitsu-eternus-dx-fc/e2857cd : FAILURE in 3m 04s","accounts_in_message":[],"_revision_number":7},{"id":"6e88a0b07ba964e5d25d40b56dc8dd3fd831674a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 07:34:38.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/61953e6a91cb4f5f8b26b9fb8a4772f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25072e8b40c1434eae00716215d05283 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a0a22317da648b496ad9f578935fb71 : SUCCESS in 8m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8e15009d9be45a199fcb22c99ffc4f6 : SUCCESS in 11m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbd603ae06fc45edb874c656fb6439d2 : SUCCESS in 13m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/4ef644239c1c4fe9b6b1e52706d11ced : SUCCESS in 52m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8020fafbe9884cf5b5527f0c47dff13f : SUCCESS in 53m 04s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b557685998ea4ee09f857568a55a740f : SUCCESS in 1h 45m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91cc4c73bcca46f0a74e46791cd322ef : SUCCESS in 42m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5915553db2314e4095f384040b54cf4b : SUCCESS in 12m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/02e1af2aedc2477c8304f85bd1598b6a : SUCCESS in 5m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cc4ae8f8399a4e11b4292f43f16140d9 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c04d95f70e5248ec8f1ad1cbead7ce80 : SUCCESS in 5m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/88cd19d283dd4c78b4e1ae295a6a5b8a : SUCCESS in 4m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/23c572611b8a49d188f75f30635783ad : SUCCESS in 47m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/805af05d792f438ebbb13b2ba767dc27 : FAILURE in 6m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/247e929ffabb4bc8a8cca215907233a0 : TIMED_OUT in 2h 04m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2399ef37081a405c81e2455cd5645c2e : FAILURE in 1h 19m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d322a641a11e4d5a9efb4ade15bb9184 : SUCCESS in 1h 22m 46s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b9522a3ef57a4529861642c853dc8a5b : SUCCESS in 53m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/60115b2c221741509264324ccc7c1dde : SUCCESS in 1h 28m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a0667c73748d48de9861cff10ae1d827 : SUCCESS in 41m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a14898a726b24b5a80467d232db6c0c7 : SUCCESS in 1h 13m 10s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/df233c44ec76433ea02aca8589b25622 : FAILURE in 2h 08m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1823bed510224b96aae08faf5050504f : SUCCESS in 50m 19s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d09f33076aea42eabe61ea4381ecdda0 : SUCCESS in 1h 31m 55s","accounts_in_message":[],"_revision_number":7},{"id":"a10cfffba483671b3f0808249258ae2d78323450","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-22 07:40:57.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/09/860209/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/c68ff6d/ : SUCCESS in 1h 37m 05s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/09/860209/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f7d315f/ : FAILURE in 2h 13m 11s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/09/860209/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/737a762/ : SUCCESS in 1h 49m 04s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/09/860209/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/40f702f/ : SUCCESS in 1h 56m 14s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/09/860209/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/16ebdbf/ : SUCCESS in 2h 11m 04s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bbf720293889f6fcd3855359cc414d3c0b6652e1","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-02-22 09:11:07.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0c59d56c0e4ec42346ab910a5cdabdc081fc3cd7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 09:23:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26a8d50a02ba430692f9f32feba83683\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/430164a7b16f4099a7d63046ac0aac69 : SUCCESS in 11m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b8c7c44b913746adaf37f885d63d193f : SUCCESS in 9m 04s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3af78acdd86c386f465c788a59d38b0a7a9946c0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-22 10:31:41.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/88d82b595a774d1bb1f6b256e7c8f78a : SUCCESS in 1h 18m 52s","accounts_in_message":[],"_revision_number":7},{"id":"4eb030e9a2596d8de63c7af54b00515834fe751d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-02-22 10:47:01.000000000","message":"Patch Set 7:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/09/860209/7/check/fujitsu-eternus-dx-iscsi/7287b01 : SUCCESS in 1h 35m 21s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/09/860209/7/check/fujitsu-eternus-dx-fc/78fa34f : FAILURE in 46m 20s","accounts_in_message":[],"_revision_number":7},{"id":"1359ce7176fb685fd983004b435888c9e03c541f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-22 11:20:55.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/09/860209/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5dc7b6d/ : SUCCESS in 1h 49m 27s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/09/860209/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0049679/ : FAILURE in 2h 06m 56s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/09/860209/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/abd37e1/ : SUCCESS in 1h 49m 44s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/09/860209/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/694f8c1/ : SUCCESS in 2h 00m 59s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/09/860209/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/1006794/ : FAILURE in 1h 57m 51s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"58e2f3749f748cdb731a1d66132dbe3489abdf2f","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-22 11:38:32.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/0335729ec26444059ee6d3cd3b8cfdc2 : SUCCESS in 1h 18m 16s","accounts_in_message":[],"_revision_number":7},{"id":"2388212f8ed22dd4604e4f43d4677ed6fd93022f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 11:40:45.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca14314cff6d44daa7d48a2f84119275\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80f63c1002cd4e7cb804c2aa532acbec : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e58ffb02f058470688869d265af22858 : SUCCESS in 9m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e246777d4b6a445dbeff96b90188786e : SUCCESS in 10m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be65b128293e4779bace1e9883b333ee : SUCCESS in 12m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/e6087eb699f3490d81bd110def01dc7e : SUCCESS in 55m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9ce42123a4244125b708993aadc25774 : SUCCESS in 45m 52s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3ff6d82d30b1409c8c6543fbce8cf898 : SUCCESS in 1h 17m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93e53919ab084d0a9b2bfe3c528ab67c : SUCCESS in 1h 06m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/23d0c2a62c534717bd946cf013186b42 : SUCCESS in 16m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8412b82a5fe347e491acefa78813b6a5 : SUCCESS in 5m 37s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/896e65bae3194a679e8a04b688f9b518 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9becf6e7ad4949cea2e2d912aa14028f : SUCCESS in 6m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a41a2a19fd174048b2f004424d211863 : SUCCESS in 5m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1247c59330af42888bf57f56f59fbe2e : SUCCESS in 37m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5168e1ee073b443ab89a962bf48514df : FAILURE in 6m 38s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/287b6512c7fa4611b825e6c56e43974f : TIMED_OUT in 2h 04m 13s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e39bca258e484b858ee324b8d826cc58 : FAILURE in 1h 33m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1c157604765b44e4a151bd193327975a : SUCCESS in 1h 58m 41s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6010feda3dd947b1ab677f2a6eefbc8a : FAILURE in 49m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2870b8784be742a0849185d75da78a51 : SUCCESS in 1h 37m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6e9766d43a0343858b9395dcc8c1d97f : SUCCESS in 46m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/96f5ca090d52471fb0bf62906ac43298 : SUCCESS in 1h 24m 47s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8ce62c097675451da64954e33ebc682b : FAILURE in 2h 26m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22f9063de532443897837ea72bdce100 : SUCCESS in 48m 11s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ed71e0075dd462dba5e7d39d3146a01 : SUCCESS in 1h 02m 33s","accounts_in_message":[],"_revision_number":7},{"id":"eb9b332a8dfd410e5ff3dae7089c268d6d8c8730","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-22 12:59:59.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/09/860209/7/check/DellEMC_Unity_FC/bc4649a/DellEMC_Unity_FC/2342 : SUCCESS in 1h 23m 22s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/09/860209/7/check/DellEMC_Unity_iSCSI/50ea1be/DellEMC_Unity_iSCSI/2196 : FAILURE in 1h 10m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"8b84bd649ebcc10d37667e8e8a2c1d6e585e0744","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-22 18:21:28.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/27f37acbbeb64e309d764411e17663d7 : SUCCESS in 2h 07m 23s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/15345ebb90fd441c932264704f1a9015 : RETRY_LIMIT in 1m 00s","accounts_in_message":[],"_revision_number":7},{"id":"d3a2ccc968095b71ff25765ffa708f2d8d8764be","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-02-23 08:41:52.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"291a779094d3213749f560d8839932815a8093dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 08:51:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c00bbbf446a74f9fa159471191c3b3a0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7af60c33d0a416a81379f9bf936b244 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7dd3137480d54e89ba7375cff370043c : SUCCESS in 8m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fe6320813495ae825a92a2722816d02a5f465845","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-23 09:10:11.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"5e367ab2ed32d086acb455fdc33df6dfc45f08d2","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-23 09:55:30.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/deeb87a8974f4a0499b66ec67cadf469 : SUCCESS in 1h 11m 51s","accounts_in_message":[],"_revision_number":7},{"id":"d462103a513066e90b285678197d02966223fc67","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-23 10:01:00.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/baa2001477734a0784d22c5326b4a266 : SUCCESS in 1h 17m 16s","accounts_in_message":[],"_revision_number":7},{"id":"0f5bfbf270a957f5a050f49ac1e311cdbcbb0d95","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-23 10:40:19.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/09/860209/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/62aa2e5/ : SUCCESS in 1h 44m 22s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/09/860209/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/dd94a2b/ : FAILURE in 1h 48m 22s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/09/860209/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/c6697bc/ : SUCCESS in 1h 49m 45s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/09/860209/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/a4bf980/ : SUCCESS in 1h 56m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c09dc3f200a488f89427b073a465afddc3d2a7c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 10:59:55.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac59c82a39314b9cb0f713dd8a00fb30\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42cc94c23641479380f926e176eeef51 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9f1b8b6af6440c48366129f7a2099d6 : SUCCESS in 10m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9674f98a447843b0978ab585e19a274d : SUCCESS in 10m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad24351245e4419d980c0f3d3a5db222 : SUCCESS in 12m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/8b2fd7887c4942c8a77bc6c32d1e3580 : SUCCESS in 54m 21s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/359f891fe8df4c65b8eabc8fe962fadd : SUCCESS in 48m 52s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3060651ab08048dd82ce7279f7424114 : FAILURE in 1h 09m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87dfad10a2c1426f97e6ebb476b753c2 : SUCCESS in 47m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/feffc4565e174c6ca1ccef0454716f11 : SUCCESS in 12m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/da3c44b9362b47f4b59304f1f2723a6b : SUCCESS in 5m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/85589b0a10724dfaa23aaac8b7f3b5b1 : SUCCESS in 18m 16s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a50482dc260345f9974c93fc95950195 : SUCCESS in 9m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8c62a996205b46a285d81a00743b8c68 : SUCCESS in 8m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/40da128e49bf41e29eafeb85630fdc17 : SUCCESS in 33m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fb534e790e894dd4ac2d91d3eaa79e44 : FAILURE in 5m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d43fa6f55790436d8f2ed6113ed69e3d : SUCCESS in 1h 08m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0e16c0447e27430ca4c031a2425af7e4 : FAILURE in 1h 27m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ede019df6ba44f83aeb4b01f870867c1 : SUCCESS in 1h 25m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/79ca75da3cf34171ac1fc16328b59f3b : SUCCESS in 1h 06m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/edd0f9a9542b40a1a1faa66e963deee6 : SUCCESS in 1h 16m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7dba167ad5ff427d913b39fb7f90ef2f : SUCCESS in 47m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/87b0dd4035ac4faea60f13ad61df494a : SUCCESS in 1h 04m 44s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d4fd9baef2634ffab6f8793f50f09be8 : SUCCESS in 2h 12m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07392047cfb44f6abf4a6e95ffd1db6b : SUCCESS in 54m 41s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77d2d10ffec04c3f8d94841f861b9f84 : SUCCESS in 1h 13m 37s","accounts_in_message":[],"_revision_number":7},{"id":"6faa70b2fa8354c227e6642fcb39a119126c154a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-02-25 06:40:57.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/09/860209/7/20230225125838/NEC-V-ISCSI : SUCCESS in 2h 16m 41s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/09/860209/7/20230225125838/NEC-V-FC : SUCCESS in 2h 42m 17s","accounts_in_message":[],"_revision_number":7},{"id":"099642b79c97fb4dc8f3053711150e55c4451ba2","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-02-28 08:49:55.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"31c1264e3f9709c0f2bbaea6a7d66699a557a6f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 09:12:25.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/948ecb1195374cc5856186dd9f3a1559\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/44f68a93ebe34a7f81032883568be454 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/478ffb861fc649a9b8251b79908b0c91 : SUCCESS in 20m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a8742419ccebb69216d5f94be117d53cd1851279","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-28 10:01:10.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7739f0076fb04d34bc23df9417fb0050 : SUCCESS in 1h 09m 20s","accounts_in_message":[],"_revision_number":7},{"id":"fc51a7050845b80183012716c999878ddad702b4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-28 10:52:28.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"b0199d509d16cacfbb00dafe0860ebeb68477a35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 10:56:22.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d04ec6f577a84a30a1f210654e7e5c28\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/962bf77f45774e1eb85d425fc37031ba : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1d4193cf69e4bd2be3ee988f499eb85 : SUCCESS in 14m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93c2744ea46a40a4a3d99e6950410d15 : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/508de9c55423493caa57e9f319e6fe65 : SUCCESS in 13m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/65d67ea2de474da1ada77e46bf0fe0aa : SUCCESS in 1h 11m 02s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ff507fb7078d43769d0a0b71694a73e2 : SUCCESS in 57m 45s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/83ed08d4a8294180a003bf4bad061e87 : SUCCESS in 1h 24m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66b0690c7f3747a793d29fd75ec2aae8 : SUCCESS in 1h 03m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ca1f18b136dc43d5a25dcd628b1bd84d : SUCCESS in 14m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/66edb1eae06444c8b444883e3e8ec054 : SUCCESS in 4m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6c8db98b38b24866bb5ef062e67f6121 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c0e77569c2bd4698bf88792181639b9b : SUCCESS in 5m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8645ff9f7b564aae8eb2472eb1c3be58 : SUCCESS in 5m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/02ec6d080ced41ec823a6f46f6e83ddd : SUCCESS in 35m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0c51bcead8d8478faca771c973365efb : FAILURE in 5m 27s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8f6ebdaaa7094eb6a67a13b6644dadaa : SUCCESS in 1h 19m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b88a6dd373f64dc19fe9c7db59804cc1 : FAILURE in 1h 19m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b3d1f6ae7cd64d198163496c436a1eef : SUCCESS in 1h 45m 37s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/31b6552ab7a9460db479e9f1d53e3558 : SUCCESS in 1h 06m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f35c9ae76dab40bd9d3ee5854df3f159 : SUCCESS in 1h 14m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/afb043b8f0334fa999a903d8b5907c59 : SUCCESS in 43m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bdd1bbb94de34b9ba423f15de200d4a3 : SUCCESS in 1h 59m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/31caa3edee00468da090dd98e525e54a : SUCCESS in 1h 54m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/234cda43abae4fc38a5f411f8dee0280 : SUCCESS in 46m 01s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14cbc0cff1a14f5c9e980d1f33766cd7 : SUCCESS in 1h 30m 57s","accounts_in_message":[],"_revision_number":7},{"id":"8ba1cec0f943869af091332c942b1f31bb169d2d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-02-28 11:04:31.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/28_02_2023_10_56_30\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"e1f8583d40eb21c1e7731a891a11ccd1a392b23d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-28 11:17:03.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/09/860209/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b30bd2f/ : SUCCESS in 1h 48m 06s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/09/860209/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/ce413be/ : FAILURE in 2h 24m 35s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/09/860209/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/f657d10/ : SUCCESS in 1h 44m 03s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/09/860209/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/a7c28b2/ : FAILURE in 2h 09m 31s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/09/860209/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/4c9d62e/ : SUCCESS in 1h 51m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"20561673aedb78bfb3394db973c06be26aa99eaa","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-28 12:09:43.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/09/860209/7/upstream-check/cinder-tempest-plugin-ontap-fcp/19fdcb0/ : SUCCESS in 1h 12m 32s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/09/860209/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/f025637/ : FAILURE in 42m 31s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/09/860209/7/upstream-check/cinder-tempest-plugin-ontap-nfs/00c466b/ : SUCCESS in 42m 02s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/09/860209/7/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/46061a0/ : SUCCESS in 42m 17s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"b33be92df42609b7194e65d555a5ba2355d980c0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-02-28 12:26:38.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/09/860209/7/check/DellEMC_PowerFlex/961b106/DellEMC_PowerFlex/1659 : SUCCESS in 1h 29m 53s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"61303c693c9a75d36019a1a63aa1ba197510a5ab","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2023-02-28 12:34:26.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/09/860209/7/check/DellEMC_SC_FC/42a97a4/DellEMC_SC_FC/1124 : SUCCESS in 1h 09m 48s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/09/860209/7/check/DellEMC_SC_iSCSI/b5f0257/DellEMC_SC_iSCSI/1108 : SUCCESS in 1h 37m 45s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"ee2597de191e131ef6651f492ca6065afcb51811","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-02-28 13:56:47.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/09/860209/7/check/DellEMC_PowerStore_iSCSI/33b1bf9/DellEMC_PowerStore_iSCSI/1053 : SUCCESS in 1h 02m 53s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/09/860209/7/check/DellEMC_PowerStore_FC/7f85fec/DellEMC_PowerStore_FC/1176 : SUCCESS in 1h 17m 23s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/09/860209/7/check/DellEMC_PowerStore_NFS/c8a54d4/DellEMC_PowerStore_NFS/692 : SUCCESS in 1h 05m 18s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/09/860209/7/check/DellEMC_PowerStore_NVMeTCP/3643757/DellEMC_PowerStore_NVMeTCP/557 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"7355d54fb96f9f23629fd9d2b7424321ccc19c79","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2023-02-28 14:22:05.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8b7d6fe07ebeb9e935584f96721734d1354e244e","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-02-28 14:49:27.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/860209-7/HPMSA-iSCSI/ : SUCCESS in 1:42:50\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/860209-7/HPMSA-FC/ : FAILURE in 1:19:29","accounts_in_message":[],"_revision_number":7},{"id":"f6d53f219453ad3bc4885184b3877014e84645c0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-28 14:50:06.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/09/860209/7/check/DellEMC_Unity_FC/1c252c2/DellEMC_Unity_FC/2376 : SUCCESS in 1h 31m 19s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/09/860209/7/check/DellEMC_Unity_iSCSI/cbd7588/DellEMC_Unity_iSCSI/2228 : SUCCESS in 1h 49m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"b7877fdfd15ad56e3be404a439c281a12c7efec7","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-02-28 15:20:47.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/860209-7/PVME-iSCSI/ : SUCCESS in 1:54:48\n- PVME-FC https://os-logs.seagate.com/dell-ci/860209-7/PVME-FC/ : SUCCESS in 1:37:58","accounts_in_message":[],"_revision_number":7},{"id":"173d8f1eacce61ed9fbbfc30fe7f7451c21e1ea8","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-02-28 15:23:03.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/860209-7/Seagate-iSCSI/ : SUCCESS in 1:58:49\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/860209-7/Seagate-FC/ : FAILURE in 1:36:27","accounts_in_message":[],"_revision_number":7},{"id":"424897f958e10d9a3bad3b92410f528ef08b70a2","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-02-28 20:44:03.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/09/860209/7/check/DellEMC_PowerMAX_FC/edbaf6f/DellEMC_PowerMAX_FC/837 : SUCCESS in 5h 56m 17s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/09/860209/7/check/DellEMC_PowerMAX_iSCSI/bd2ec46/DellEMC_PowerMAX_iSCSI/817 : FAILURE in 6h 21m 35s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":7},{"id":"633d60e099df1375ab018db842c7a17601a0d0ce","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-03-01 03:28:23.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a9426625ef077cf11b7bb5036eecaa792299d98d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-01 08:08:20.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"21ce75d9856b39b07eddedd6217baa6b65966f78","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-03-02 04:10:17.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/09/860209/7/20230302093754/NEC-FC : SUCCESS in 3h 30m 26s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/09/860209/7/20230302093754/NEC-ISCSI : SUCCESS in 3h 32m 20s","accounts_in_message":[],"_revision_number":7},{"id":"b0e0dacb9b690c70126e05a55ae7bc3676c63994","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-03-02 09:08:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e044729357b54d0e4e11e93f44dfbc9b24dcebc8","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-03-03 11:32:48.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"212201648a21a8777d953fe2aa83e65e7818792b","author":{"_account_id":32465,"name":"IBM Storage CI","email":"hchintha@in.ibm.com","username":"ibm-svf-user"},"date":"2023-03-03 12:59:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- storwize-tempest-dsvm-full-iscsi https://ansiblecinder-logs.s3.amazonaws.com/storwize-tempest-dsvm-full-iscsi/f2ae016/index.html : SUCCESS in 1h 25m 52s\n- storwize-tempest-dsvm-full-fc https://ansiblecinder-logs.s3.amazonaws.com/storwize-tempest-dsvm-full-fc/a5c7e55/index.html : SUCCESS in 51m 10s","accounts_in_message":[],"_revision_number":7},{"id":"fc2d336d10b2db43ab931a4daa330178f129c9af","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-03-03 17:41:00.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6c91edb28036831d4885faf1493016087594f045","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 17:41:51.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"6df8a4c13049714570af2421b3af0f11583058a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 20:47:43.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5eb3dae163a74b0ba0b14b22331ac798\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76170faff7354070bcc04cf0610bdcfa : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26fb9a7357134847b0438fab180e4fd1 : SUCCESS in 8m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f04c843b90a543429957182c1b8bd6c9 : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e74263a4904740f8ad9886cff571d487 : SUCCESS in 13m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/91dec0cf517841f4be16c53fe6df05ee : SUCCESS in 50m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c34ada05dfea4639992b6edbeef1428c : SUCCESS in 1h 38m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04902778fa884fc2a36c2e280191780a : SUCCESS in 46m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/165480c6c157458ea0f07d739d2051a7 : SUCCESS in 1h 24m 12s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/52b0e63102514241997c57df1bfb7c54 : SUCCESS in 1h 00m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/001280b6506c42be891e165c9b590989 : SUCCESS in 1h 09m 05s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/115175e4e321400b9979b06673fbc58b : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":7},{"id":"b3df012307aa3c162bfb47331123f7b26815c25b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 20:47:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"30fb8da68ad75144f9cad7d7101e3233ba516de6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 20:48:58.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a257a3cd75e9449992279b4b33bddb8c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3fc5b45144a49dca5c74cfd6cebfb6c : SUCCESS in 54s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"e36d64585b9a250d6dd53392c3350004b33fdbb6","revisions":{"c1050b6d37cc4659a22ccb8c2c933ab5adf3af4a":{"kind":"REWORK","_number":1,"created":"2022-10-04 07:52:44.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/09/860209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/09/860209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/09/860209/1"}}},"commit":{"parents":[{"commit":"c94962b98e599a35e5c00bdbeb699edee19bf33e","subject":"Merge \"setup.cfg: [extras] for pip-installable driver requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c94962b98e599a35e5c00bdbeb699edee19bf33e"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-10-04 07:41:29.000000000","tz":330},"committer":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-10-04 07:52:38.000000000","tz":330},"subject":"Update IBM Storwize drivers","message":"Update IBM Storwize drivers\n\nIBM Storwize drivers are currently reporting\nallocated_capacity_gb. Drivers should not report\nthis and let Cinder core take care of this value.\n\nCloses-Bug: #1746223\nChange-Id: I1c1600edb7c1323e8ad7396b3e5dc4652b04f9f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1050b6d37cc4659a22ccb8c2c933ab5adf3af4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c1050b6d37cc4659a22ccb8c2c933ab5adf3af4a"}]},"branch":"refs/heads/master"},"18e1c98446133513013468fec54c000444e71038":{"kind":"REWORK","_number":2,"created":"2022-10-06 10:24:29.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/09/860209/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/09/860209/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/09/860209/2"}}},"commit":{"parents":[{"commit":"c94962b98e599a35e5c00bdbeb699edee19bf33e","subject":"Merge \"setup.cfg: [extras] for pip-installable driver requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c94962b98e599a35e5c00bdbeb699edee19bf33e"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-10-04 07:41:29.000000000","tz":330},"committer":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-10-06 10:23:18.000000000","tz":330},"subject":"Update IBM Storwize drivers","message":"Update IBM Storwize drivers\n\nIBM Storwize drivers are currently reporting\nallocated_capacity_gb. Drivers should not report\nthis and let Cinder core take care of this value.\n\nCloses-Bug: #1746223\nChange-Id: I1c1600edb7c1323e8ad7396b3e5dc4652b04f9f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/18e1c98446133513013468fec54c000444e71038"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/18e1c98446133513013468fec54c000444e71038"}]},"branch":"refs/heads/master"},"22252485fb88e61d3f851e4cc2252f3a10edf3d3":{"kind":"REWORK","_number":3,"created":"2022-10-10 06:03:55.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/09/860209/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/09/860209/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/09/860209/3"}}},"commit":{"parents":[{"commit":"c94962b98e599a35e5c00bdbeb699edee19bf33e","subject":"Merge \"setup.cfg: [extras] for pip-installable driver requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c94962b98e599a35e5c00bdbeb699edee19bf33e"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-10-04 07:41:29.000000000","tz":330},"committer":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-10-10 06:03:07.000000000","tz":330},"subject":"Update IBM Storwize drivers","message":"Update IBM Storwize drivers\n\nIBM Storwize drivers are currently reporting\nallocated_capacity_gb. Drivers should not report\nthis and let Cinder core take care of this value.\n\nCloses-Bug: #1746223\nChange-Id: I1c1600edb7c1323e8ad7396b3e5dc4652b04f9f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/22252485fb88e61d3f851e4cc2252f3a10edf3d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/22252485fb88e61d3f851e4cc2252f3a10edf3d3"}]},"branch":"refs/heads/master"},"78dee4e6d5830d47c142f9416fd5d361c4b1f4fe":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-10-10 11:11:12.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/09/860209/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/09/860209/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/09/860209/4"}}},"commit":{"parents":[{"commit":"4a6792fba3c716a4af6443eeaa688b6f32a60c08","subject":"Merge \"Pure Storage - bump version numbers for Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4a6792fba3c716a4af6443eeaa688b6f32a60c08"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-10-04 07:41:29.000000000","tz":330},"committer":{"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","date":"2022-10-10 11:11:12.000000000","tz":0},"subject":"Update IBM Storwize drivers","message":"Update IBM Storwize drivers\n\nIBM Storwize drivers are currently reporting\nallocated_capacity_gb. Drivers should not report\nthis and let Cinder core take care of this value.\n\nCloses-Bug: #1746223\nChange-Id: I1c1600edb7c1323e8ad7396b3e5dc4652b04f9f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/78dee4e6d5830d47c142f9416fd5d361c4b1f4fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/78dee4e6d5830d47c142f9416fd5d361c4b1f4fe"}]},"branch":"refs/heads/master","description":"Rebase"},"8f236b711c6f9e6f0c5520a1d8ecdfafc98d59fa":{"kind":"REWORK","_number":5,"created":"2022-10-17 09:42:27.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/09/860209/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/09/860209/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/09/860209/5"}}},"commit":{"parents":[{"commit":"4a6792fba3c716a4af6443eeaa688b6f32a60c08","subject":"Merge \"Pure Storage - bump version numbers for Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4a6792fba3c716a4af6443eeaa688b6f32a60c08"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-10-04 07:41:29.000000000","tz":330},"committer":{"name":"Tushar Gite","email":"tushargite96@gmail.com","date":"2022-10-17 09:42:19.000000000","tz":330},"subject":"Update IBM Storwize drivers","message":"Update IBM Storwize drivers\n\nIBM Storwize drivers are currently reporting\nallocated_capacity_gb. Drivers should not report\nthis and let Cinder core take care of this value.\n\nCloses-Bug: #1746223\nChange-Id: I1c1600edb7c1323e8ad7396b3e5dc4652b04f9f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f236b711c6f9e6f0c5520a1d8ecdfafc98d59fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8f236b711c6f9e6f0c5520a1d8ecdfafc98d59fa"}]},"branch":"refs/heads/master"},"734acbcc587c6f13e2954be7f3f39a553015735d":{"kind":"REWORK","_number":6,"created":"2022-11-01 16:40:53.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/09/860209/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/09/860209/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/09/860209/6"}}},"commit":{"parents":[{"commit":"4a6792fba3c716a4af6443eeaa688b6f32a60c08","subject":"Merge \"Pure Storage - bump version numbers for Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4a6792fba3c716a4af6443eeaa688b6f32a60c08"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-10-04 07:41:29.000000000","tz":330},"committer":{"name":"Tushar Gite","email":"tushargite96@gmail.com","date":"2022-11-01 16:35:08.000000000","tz":330},"subject":"Update IBM Storwize drivers","message":"Update IBM Storwize drivers\n\nIBM Storwize drivers are currently reporting\nallocated_capacity_gb. Drivers should not report\nthis and let Cinder core take care of this value.\n\nCloses-Bug: #1746223\nChange-Id: I1c1600edb7c1323e8ad7396b3e5dc4652b04f9f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/734acbcc587c6f13e2954be7f3f39a553015735d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/734acbcc587c6f13e2954be7f3f39a553015735d"}]},"branch":"refs/heads/master"},"e36d64585b9a250d6dd53392c3350004b33fdbb6":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-02-22 05:25:02.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/09/860209/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/09/860209/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/09/860209/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/09/860209/7"}}},"commit":{"parents":[{"commit":"a92aa06e463f6322e8abdc2f570a801ea82dd1cb","subject":"Hitachi: add GAD volume support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a92aa06e463f6322e8abdc2f570a801ea82dd1cb"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-10-04 07:41:29.000000000","tz":330},"committer":{"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","date":"2023-02-22 05:25:02.000000000","tz":0},"subject":"Update IBM Storwize drivers","message":"Update IBM Storwize drivers\n\nIBM Storwize drivers are currently reporting\nallocated_capacity_gb. Drivers should not report\nthis and let Cinder core take care of this value.\n\nCloses-Bug: #1746223\nChange-Id: I1c1600edb7c1323e8ad7396b3e5dc4652b04f9f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e36d64585b9a250d6dd53392c3350004b33fdbb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e36d64585b9a250d6dd53392c3350004b33fdbb6"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
