)]}'
{"id":"openstack%2Fcinder~860212","triplet_id":"openstack%2Fcinder~master~I62c3d41ef92e0d474f5315b5bb1186faa187d958","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-12-01 14:15:36.000000000","reason":"Change was submitted"},"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2022-10-07 11:34:14.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"30615":{"account":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"last_update":"2022-12-01 14:15:36.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2022-10-07 10:44:15.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2022-12-01 14:15:36.000000000","reason":"Change was submitted"},"7198":{"account":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"last_update":"2022-11-30 14:14:31.000000000","reason":"\u003cGERRIT_ACCOUNT_7198\u003e replied on the change","reason_account":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-12-01 14:15:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I62c3d41ef92e0d474f5315b5bb1186faa187d958","subject":"Toyou: Remove allocated_capacity_gb calculation","status":"MERGED","created":"2022-10-04 08:02:35.000000000","updated":"2022-12-01 14:35:11.000000000","submitted":"2022-12-01 14:15:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860212","meta_rev_id":"6eaa652a302a2f1b14d9c1b0119f7641556bb1d9","_number":860212,"virtual_id_number":860212,"owner":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-01 14:15:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"recommended":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"all":[{"value":1,"date":"2022-10-21 09:44:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":1,"date":"2022-11-04 14:39:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2022-10-07 11:34:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":1,"date":"2022-11-10 16:39:34.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2022-10-21 12:27:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-30 06:12:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":2,"date":"2022-11-16 13:29:49.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":2,"date":"2022-11-30 14:14:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":1,"date":"2022-11-30 14:14:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-04 08:29:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-04 09:07:23.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-10-04 09:24:59.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-10-04 09:39:24.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-04 09:55:35.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-04 10:14:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-04 11:02:15.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-10-04 11:02:23.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2022-10-04 11:05:41.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-10-04 11:34:25.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-10-04 11:50:29.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-04 16:45:46.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2022-10-04 20:03:04.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-10-04 21:15:14.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-10-05 16:41:54.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-06 09:34:39.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-10-06 10:09:32.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-10-06 10:13:25.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-06 10:21:05.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-10-06 10:38:05.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-10-06 11:01:07.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-10-07 10:44:15.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2022-10-07 11:04:03.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-07 11:34:14.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2022-10-07 13:08:11.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-10-11 09:56:09.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2022-10-11 09:56:09.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-10-20 09:44:16.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-10-21 09:44:02.000000000","updated_by":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"reviewer":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"state":"REVIEWER"},{"updated":"2022-10-21 12:27:42.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"state":"REVIEWER"},{"updated":"2022-10-30 06:12:00.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-11-02 14:05:16.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-11-04 14:39:56.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2022-11-10 16:39:34.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2022-11-16 13:29:49.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2022-11-30 11:23:56.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"}],"messages":[{"id":"df63568660215639de0efe82d26522ebd525b23a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-04 08:02:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86ac0f568634b7b0e35cf2e8ed2675c0da79b168","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 08:29:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b148ebaf37249a498e490d894fc70d2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f4d6a52c80cb41a48df00e5af47574a4 : SUCCESS in 22m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3e72f814546845e986da6ae2b90743d0 : SUCCESS in 20m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3652d889757bed998b90ec77bf2b85e3b97a9660","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-10-04 09:07:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/860212/1/iscsi/ : SUCCESS in 1h 04m 15s\n- cinder-smb http://cloudbase-ci.com/cinder/860212/1/smb/ : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":1},{"id":"21bfe9b2307714b79b62829d026b92e43481dff8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-10-04 09:24:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/12/860212/1/check/iscsi-driver-alletra-6k/d584687/d584687.html : FAILURE in 1h 17m 29s","accounts_in_message":[],"_revision_number":1},{"id":"dc0dfc7c31f21b0b03883e907f36eb15551840ec","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-04 09:39:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/860212 : FAILURE in 1h 35m 10s","accounts_in_message":[],"_revision_number":1},{"id":"54c468bc837dd2927874235d9e07cbd4012d9ab2","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-10-04 09:55:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/12/860212/1/ : FAILURE in 55m 36s","accounts_in_message":[],"_revision_number":1},{"id":"57fb9e66c9aaa5bb8c9724603643bc943b7e7bd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 10:14:04.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a37d30a3109d4c4f8c50553108115499\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7a5cbf27205486e8179e81f38b7b383 : FAILURE in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8baa9ce99e3b4dddb5604f6f1608ce71 : SUCCESS in 12m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9204a49c51148eb9483e6a312ac6b14 : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4d4021b9bea4bf88c589f34b9c4858c : SUCCESS in 15m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/088763056b294dd8bbf1c6b79b5343e1 : SUCCESS in 56m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b50ff93fde6447ed8b88e71241a44bd1 : SUCCESS in 51m 57s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/328e632503424e2294eea71a99e08b3e : SUCCESS in 1h 13m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2ba5a2b88754a21941b5f40dfaf6d4c : SUCCESS in 50m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c10395b1d8f043daa276f4f66e8319fe : SUCCESS in 16m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5fd1a21624244f309b563e2305f4491a : SUCCESS in 5m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f802129190024b34b36d6a205144dddb : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/faad4162537f4300954382f3bf0e3b04 : SUCCESS in 6m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/31b3abb6606f45faa4fcec98efd3b199 : SUCCESS in 9m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0e77c88e533f48949543c3500685a592 : SUCCESS in 32m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/86f194073ddc41dcbc96f4ea660a8fcf : FAILURE in 8m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cad9a1a48e434d9eb851a6c013f95217 : SUCCESS in 1h 04m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7689cceb388240d89ebde8316eaec309 : FAILURE in 1h 17m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/98b412b1c536466d9ddec793cc12ecc5 : SUCCESS in 1h 32m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7bbd58b4e3054c5fa69997b73a835768 : SUCCESS in 58m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/13581b8b5654464e903b6c75648faf0f : FAILURE in 1h 39m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/922eaf8b9dc04e1c9ab9769827e4121b : SUCCESS in 38m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8bd3c02a3b7446d3945fc608b271e527 : SUCCESS in 1h 25m 41s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aa13821fee1e481081e7dd4e37cbb99d : SUCCESS in 1h 54m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/168029876ba1480a92adf71606c63406 : SUCCESS in 42m 14s","accounts_in_message":[],"_revision_number":1},{"id":"301b3bef839a7ad5521b7361b1cdb228f20eddd9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-04 11:02:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/12/860212/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/f32b18f/ : SUCCESS in 1h 25m 54s","accounts_in_message":[],"_revision_number":1},{"id":"d683ed356f560aebafec73e974c7ae57afb087ba","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-10-04 11:02:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/ab49750a0a374190963d24d873d83e45 : SUCCESS in 1h 07m 56s","accounts_in_message":[],"_revision_number":1},{"id":"d108736de91d557ca2d9e01113ba4a03df07b190","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-04 11:05:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e5b568ba769fb67a6ed9cce26d405d5c212e0403","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-10-04 11:05:41.000000000","message":"Patch Set 1:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":1},{"id":"3d3c3808511ee75799db8f104f876aad285c1d5a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-04 11:07:01.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"024660be032bdf2140d494e2f0d0f701a2777714","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-10-04 11:34:25.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1664870562886 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"607706194c0778b22a3c1485654afc5dc4a34a29","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-10-04 11:50:29.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-12-860212-1/ns5_nfs : SUCCESS in 3h 32m 41s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-12-860212-1/ns5_iscsi : SUCCESS in 3h 34m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"bcd1b99541b67bdd9d6b4791d24fbc4231d6087f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 11:54:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15870ccfff5b471f9dff9ec1282634bb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd79416c202042b08db447e64a7c197a : FAILURE in 38m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9f9f36b19667415dac06047cab7bdfe0 : FAILURE in 37m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"71f72cddb41be10747d8a6d3b43c8ac77a34a3ac","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-04 12:17:07.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/860212 : FAILURE in 1h 10m 06s","accounts_in_message":[],"_revision_number":2},{"id":"453419a3792be15f5305399a904bc020b9a6b894","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-10-04 12:20:03.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Cinder Windows CI, add a review comment with run-Cloudbase Cinder Windows CI\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/860212/2/iscsi/ : FAILURE in 1h 13m 49s\n- cinder-smb http://cloudbase-ci.com/cinder/860212/2/smb/ : SUCCESS in 1h 06m 11s","accounts_in_message":[],"_revision_number":2},{"id":"958f09652144818038eadc3258d9983cba578559","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-10-04 12:49:44.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/12/860212/2/check/iscsi-driver-alletra-6k/b4b1303/b4b1303.html : FAILURE in 1h 28m 31s","accounts_in_message":[],"_revision_number":2},{"id":"957f9701f01821e25ac6ba921c9c91f6efc119c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 13:32:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7994b9183724ca4827b3ba112118170\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e2453e02bfe4af39b9fd88b91a46559 : SUCCESS in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54e3cab5333b4b4695ad51ffa95df55d : SUCCESS in 10m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af0b3c6b7d51408b854896b6d38de8f2 : SUCCESS in 12m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9c000ca13074d96842ee5552a6440d3 : SUCCESS in 13m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/566c169d8b034aed8597114aed140320 : FAILURE in 14m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2fe90da6f8224821bf890a93aedc9612 : SUCCESS in 47m 57s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2f1bb4df429c431da4509667942b810d : SUCCESS in 1h 12m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca00ee3a0fc04644a6c95eb1ea743c80 : SUCCESS in 43m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5d27cceeef5d41af8d346dcd967cc891 : SUCCESS in 16m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/05d69590bc514fe4afbf02225957f124 : SUCCESS in 5m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6e12f93c5e68452487cb0a8c1984187e : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e60ba921827049e9929c7340654b2ae7 : SUCCESS in 6m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8803927bb19648a198ebad14d4a45df4 : SUCCESS in 5m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d85f2b4104e74457929c268ee22a5c8f : SUCCESS in 36m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2c75d63dde6443058698c3cb3bd13dd1 : FAILURE in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ce3acabd67e14988a7dcd61fddd14509 : SUCCESS in 1h 13m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c2a8e6cfb1034f30af44ab02e196513e : FAILURE in 1h 33m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/437cbaae8704420d8346416e7a9a4aad : SUCCESS in 1h 54m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5549f4298b44449b985bb882a3fd471a : SUCCESS in 54m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f252bb2162914897a0b1a97793854072 : SUCCESS in 1h 31m 53s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c7b7fdd45c4d4a4c858782b9323dd713 : SUCCESS in 32m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/17ef93a4716a44adae3c988379ff3ce6 : SUCCESS in 1h 30m 11s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5437b06c30134a079f2a47b2922fda78 : SUCCESS in 2h 19m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a40d4715a9846ee8008ca303469ed01 : SUCCESS in 1h 01m 24s","accounts_in_message":[],"_revision_number":2},{"id":"199104be2b19136741c58fe13c5a48fc01d2f621","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-04 14:02:31.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/860212 : FAILURE in 1h 36m 41s","accounts_in_message":[],"_revision_number":2},{"id":"896be063c98f3ff4cede4fe1c1a576f505316261","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-10-04 14:37:55.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-12-860212-2/ns5_nfs : SUCCESS in 3h 31m 0s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-12-860212-2/ns5_iscsi : SUCCESS in 3h 31m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"4180b0f2a3e09b72cb5a3bf65cff8c4257c87d1b","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-10-04 15:01:35.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/12/860212/2/ : FAILURE in 53m 14s","accounts_in_message":[],"_revision_number":2},{"id":"d6d5901f05c1039d298950d340c21177cacaddca","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-10-04 15:17:32.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/38201dd8623a420ba1863e0436b2c1cb : FAILURE in 1h 06m 56s","accounts_in_message":[],"_revision_number":2},{"id":"18a4bb3f8f8eeadf09c5ac99770bf1bc7e8f22f9","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-10-04 16:45:46.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ec890129d7c96999b1d8c8400934a69b1add942e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-10-04 20:03:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/12/860212/2/check/dsvm-tempest-cisco-zonemanager-job/7ed2352/ : SUCCESS in 1h 19m 56s","accounts_in_message":[],"_revision_number":2},{"id":"4a7b4f88d685c1cb1327a0232f83a736ebb31033","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-10-04 20:14:01.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1664881548570 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"f5ffde56cba884c075033aaeb9134c7fdce1d367","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-04 21:15:14.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/12/860212/2/check/DellEMC_Unity_FC/1680804/DellEMC_Unity_FC/1898 : SUCCESS in 1h 10m 54s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/12/860212/2/check/DellEMC_Unity_iSCSI/99cbaf9/DellEMC_Unity_iSCSI/1745 : SUCCESS in 1h 52m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"aecf79c8a7312c5ab99a566fecda02f8cb3be774","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-10-05 16:41:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/12/860212/2/20221005220314/NEC-FC : SUCCESS in 3h 34m 23s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/12/860212/2/20221005220314/NEC-ISCSI : SUCCESS in 3h 38m 37s","accounts_in_message":[],"_revision_number":2},{"id":"10b94615ba48044901d6bdd2650f641be00d9e1a","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-06 03:51:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dada1a2d1d38ee09240bc648f47da64f7849e3a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-06 07:20:07.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"56fbaefc5de32bbbe8cd3010245d124d9d042445","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-06 07:20:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d8f1c9f6e9ace2492bd0bcb735bcc9ced995f33a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-10-06 07:31:37.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 8m 17s","accounts_in_message":[],"_revision_number":3},{"id":"3a07d4e4facd680833cf2c1c9c9afe6be10475eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 07:39:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70f353f0b6c44b908206eba172b28619\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/428afb964a4948ba81172af61d6cc7a7 : SUCCESS in 13m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d6333a9a0b7942edb576af8b7785e367 : SUCCESS in 13m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5afc650a46c5718f3b3a49c339291602604f8920","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-10-06 07:49:26.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1665040815375 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"410fcf3ca00d1eb02bb5bc483d932b1b8533e559","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-10-06 08:21:55.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/54f0574fc0bf4f0db96512c003754255 : SUCCESS in 1h 01m 09s","accounts_in_message":[],"_revision_number":3},{"id":"7bc2ccf46c1f4c2efc7551688e97b5cad7e463c9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-10-06 08:25:09.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/12/860212/3/check/iscsi-driver-alletra-6k/7d9b2f5/7d9b2f5.html : FAILURE in 1h 03m 50s","accounts_in_message":[],"_revision_number":3},{"id":"d99614e4b67667c80bd03cf0afd910ae131edf29","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-10-06 08:25:43.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/12/860212/3/ : FAILURE in 57m 34s","accounts_in_message":[],"_revision_number":3},{"id":"6add0126697f7aba0d8be28e9c4b44d6337d24b9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-06 08:32:49.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/860212 : FAILURE in 1h 11m 03s","accounts_in_message":[],"_revision_number":3},{"id":"bbc3f24d4788374ea342446088f4b1135485f5d6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-10-06 08:34:50.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/860212/3/iscsi/ : SUCCESS in 1h 14m 10s\n- cinder-smb http://cloudbase-ci.com/cinder/860212/3/smb/ : SUCCESS in 53m 53s","accounts_in_message":[],"_revision_number":3},{"id":"e378a83bb65dd2b09c4cb3dcef76d10b106380c5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-06 08:56:15.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/860212 : SUCCESS in 1h 34m 15s","accounts_in_message":[],"_revision_number":3},{"id":"dc84911b92e349176bc302b1da10fa1add59568b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 09:16:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/739c46d3baaa4895bb432305708634bc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acf55255ca3a4129b1dee025a1a58ddb : SUCCESS in 8m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e5605ad2b1f48f2aad5eae63ba9f8ee : SUCCESS in 12m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/299c574631654446a1655f70421e7085 : SUCCESS in 14m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02cb4d54e06f44c6a121825aff9d689a : SUCCESS in 12m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/f6c2371773f5495b876edb9d20d147a7 : SUCCESS in 59m 05s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8baf0b5581fd4fb7a01249872e3c8af8 : SUCCESS in 55m 02s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5225ebd750504999942b40d6a4219bd7 : SUCCESS in 1h 21m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dda1d2493eac4eec9d58b95194822570 : SUCCESS in 56m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9c0714ad20dd489db5725e993ba0b1b2 : SUCCESS in 14m 00s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a826bfe4fe5c43c7bf4419736a0fc860 : SUCCESS in 5m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fa21130239a54220a668fa037a24c97e : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fcb6739d7b4c425d976b032145d35ac3 : SUCCESS in 8m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b711a5a4d106457e89c06c5e04d71a54 : SUCCESS in 7m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c61cf2b5a1f8471581944dc60d2e4a02 : SUCCESS in 46m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a79d27eba3434914b8d62bee27bed5e3 : FAILURE in 9m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a9a1e4421a774496abb06bfcfd7f0dbf : SUCCESS in 1h 12m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a90709e948274d54ba9da33e95d0a5cc : FAILURE in 42m 02s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/df6e78f02c014655932eed993458fac5 : SUCCESS in 1h 35m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d685d906e76a4e13a92cb810c4dffc92 : SUCCESS in 1h 03m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/856570d71af1402e9bf4fb5ce7d73836 : SUCCESS in 1h 12m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e45a78d3337b4690b3b467ae9eafdc9f : SUCCESS in 48m 36s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/630616624f694c57badbe0291a8421da : SUCCESS in 1h 24m 38s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/baea883be4174662a4168cff8b35a63b : SUCCESS in 1h 53m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fa45df8bf244e24b87cf11a751883f1 : SUCCESS in 58m 50s","accounts_in_message":[],"_revision_number":3},{"id":"0a219e5753815f211acd24899662c772df3667d2","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-10-06 09:34:39.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-12-860212-3/1665047792 : SUCCESS in 18m 05s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":3},{"id":"6daa827aea3b1cf99a60a3c58f65f3cd4e68b82e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-06 09:43:01.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/12/860212/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/1d35aee/ : SUCCESS in 1h 48m 08s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/12/860212/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/81bd177/ : SUCCESS in 1h 43m 42s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/12/860212/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/9f003da/ : FAILURE in 2h 21m 42s","accounts_in_message":[],"_revision_number":3},{"id":"f5bc69e1da497284e7c078a3d5244c7cbe92455a","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-10-06 10:09:32.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/12/860212/3/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/439af4e/ : SUCCESS in 51m 44s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"a6e8c063a2497d5087373cb14e1e0fc42d607d67","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-06 10:10:15.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/12/860212/3/check/DellEMC_Unity_FC/80b5776/DellEMC_Unity_FC/1906 : SUCCESS in 1h 17m 04s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/12/860212/3/check/DellEMC_Unity_iSCSI/ea3bb37/DellEMC_Unity_iSCSI/1753 : SUCCESS in 1h 54m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"efa70a39043920ec8514aac42dc17c2a27a1def2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-10-06 10:13:25.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/12/860212/3/check/DellEMC_PowerFlex/54cf6c9/DellEMC_PowerFlex/1402 : SUCCESS in 56m 34s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"7cb524406aeedac6cf5b71f34950d08b3cd0f5dc","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-10-06 10:21:05.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/12/860212/3/upstream-check/cinder-tempest-plugin-ontap-fcp/4f96492/ : SUCCESS in 1h 03m 40s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/12/860212/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/a857e4c/ : SUCCESS in 45m 07s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/12/860212/3/upstream-check/cinder-tempest-plugin-ontap-nfs/e49ef6d/ : SUCCESS in 43m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"f57e90b8e027c4b9c882994b851470fa4472cade","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-10-06 10:38:05.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/12/860212/3/check/DellEMC_SC_FC/854bfc0/DellEMC_SC_FC/1048 : SUCCESS in 50m 04s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/12/860212/3/check/DellEMC_SC_iSCSI/6981653/DellEMC_SC_iSCSI/1032 : SUCCESS in 1h 21m 14s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":3},{"id":"918d5c55ea683ec73e0bd153f9c7981f8700f0bf","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-10-06 10:50:56.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-12-860212-3/ns5_nfs : FAILURE in 3h 27m 26s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-12-860212-3/ns5_iscsi : SUCCESS in 3h 30m 16s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"087219f6d192634aa348d397c358a2d5f30b56bc","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-10-06 11:01:07.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/12/860212/3/check/DellEMC_PowerStore_iSCSI/f402168/DellEMC_PowerStore_iSCSI/832 : SUCCESS in 1h 33m 52s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/12/860212/3/check/DellEMC_PowerStore_FC/a2cedf4/DellEMC_PowerStore_FC/984 : SUCCESS in 1h 01m 44s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/12/860212/3/check/DellEMC_PowerStore_NFS/4491bb1/DellEMC_PowerStore_NFS/523 : SUCCESS in 40m 55s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/12/860212/3/check/DellEMC_PowerStore_NVMeTCP/35280a3/DellEMC_PowerStore_NVMeTCP/394 : FAILURE in 1h 44m 13s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"401ec68864d652ba04d2bd2027133b8684780685","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-10-06 17:00:41.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/12/860212/3/20221006191523/NEC-FC : SUCCESS in 3h 30m 52s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/12/860212/3/20221006191523/NEC-ISCSI : SUCCESS in 6h 45m 14s","accounts_in_message":[],"_revision_number":3},{"id":"f6d6b38bb5f1e9f01b46c7f1d8b63f96f98e55be","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-10-07 10:44:15.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f6305622d63f5706d7dc164652a644b6569fb4cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-07 10:51:07.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"a29cd7e363a26877f9889b989e2e3b35c1733494","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-07 10:54:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"07d1cc752270b9127a13d46f5f31a692b2d2fe61","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-07 11:04:03.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/38af757fe1774daabd374cabfd7202d1 : RETRY_LIMIT in 22s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/8db704c14ed940009661051dfef9ed0c : RETRY_LIMIT in 20s","accounts_in_message":[],"_revision_number":4},{"id":"c9253aa92c73d5373e38945308d4b2dbd950d6b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 11:10:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3825e8c16bf949f79aac7ef1861a150c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/551dc3f8025c4b42be1ff1e99cb61fba : SUCCESS in 13m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9cc815e0bad84de9b302aee750ae7592 : SUCCESS in 13m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"261aa96d8c737f0ba72f76d10b1a813ae2d28ebe","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-07 11:23:49.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/60d112e03d68476eacfdfccd24904a98 : RETRY_LIMIT in 49s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/0d87b6b8db054c0192c8caeddeea2674 : RETRY_LIMIT in 26s","accounts_in_message":[],"_revision_number":4},{"id":"72db5ff2f7fa9c4bda31eb0b42765c4f1bdec4f8","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2022-10-07 11:34:14.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e2ccb9bf8668455885db6978b9637acecf11659f","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-10-07 11:48:54.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/017ecadda32a4f9c87b21230c474d938 : SUCCESS in 57m 02s","accounts_in_message":[],"_revision_number":4},{"id":"54c793f6684ceba37432c970f9047f09c888def6","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-10-07 11:54:43.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/12/860212/4/ : FAILURE in 55m 09s","accounts_in_message":[],"_revision_number":4},{"id":"649c8729eb6165941d9a015781ad46565f539f56","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-10-07 11:55:58.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/12/860212/4/check/iscsi-driver-alletra-6k/16557b9/16557b9.html : FAILURE in 1h 03m 48s","accounts_in_message":[],"_revision_number":4},{"id":"35b8dc9880275b7c43e4afe3e94f67ca2c6eab54","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-10-07 12:33:16.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1665139878366 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"371db1466f2ac5414ec58f0709566baed7c36bd0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-10-07 12:37:40.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/12/860212/4/check/dsvm-tempest-cisco-zonemanager-job/76c56e6/ : FAILURE in 1h 18m 15s","accounts_in_message":[],"_revision_number":4},{"id":"044591a8cc4019f09afbda6b90134b84f24206de","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-07 12:45:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/860212 : SUCCESS in 1h 52m 55s","accounts_in_message":[],"_revision_number":4},{"id":"915d49c966ceede05f9d8ad15775e9348e5bb806","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-07 12:56:23.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/74f2f06bc9574b7faa154f0ce99ccf68 : RETRY_LIMIT in 24s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/e8bb78a855d6492d82c444ab6be48036 : RETRY_LIMIT in 21s","accounts_in_message":[],"_revision_number":4},{"id":"2c125d2b9e338893bc31f10a044b4422bb3bf327","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-07 13:00:39.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/12/860212/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/0c37ab0/ : SUCCESS in 1h 44m 45s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/12/860212/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a00a913/ : SUCCESS in 1h 34m 21s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/12/860212/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/d58196d/ : FAILURE in 2h 08m 21s","accounts_in_message":[],"_revision_number":4},{"id":"04d49790dbb262f8d2907b2d463e1c90d08fc87e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 13:00:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bb0d0b9f9e945fe85fc8f9d0cd7e7bf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d05f1fb935484729a2d0be648ed99a60 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7edf9ab3635841d180fcc19fdf5bb4c1 : SUCCESS in 13m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/726e097820d94671af6e8fdf00607558 : SUCCESS in 12m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3e38551e09c40b7876ee44ffdf1cd5a : SUCCESS in 12m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/9c0b9f06150d408f8b252c586327a2cf : SUCCESS in 1h 01m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a3c3c3285aca46fdb1e95f4742afe843 : SUCCESS in 1h 00m 34s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/293d13d64ee64f9eb868b16881ae1d8b : SUCCESS in 1h 18m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75b4f1a2d5ad4646b8adfe2fc44e92ef : SUCCESS in 53m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/65461a49ca304399ac9720a3516db1e3 : SUCCESS in 13m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f3c6cf95142a42b3816a99f2db71b3a1 : SUCCESS in 5m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dba69f3d90814bc78d41674bdbf4c0d3 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1abaeb71ea164947a0706d9888411a1e : SUCCESS in 5m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/629e745263de496bbf784400380e3869 : SUCCESS in 6m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/43b1974973604b43b67ed21f9f772551 : SUCCESS in 46m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0cc7ca8909214ebb9c3ecaad4513bd37 : FAILURE in 6m 00s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/edfe2d39deed471790b591aed079c459 : SUCCESS in 1h 11m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4eb1ec633266433ba7b02b69699100d3 : FAILURE in 1h 19m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9bb0deed084343ba80544fdeca553037 : SUCCESS in 1h 59m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4a31b9e72de64fe6b6fc0bea2737c19e : SUCCESS in 1h 00m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4b8989d4291545ce8de392d00ae3e2e5 : SUCCESS in 1h 29m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fd98bbfe29c24c1f8316111307ca9c2e : SUCCESS in 40m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/578dafb677b6428eade406a39482ab2d : SUCCESS in 1h 13m 48s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/233247b890df43b6af416dbb8e3d0fe6 : SUCCESS in 2h 04m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bde5c51c6e174eb49d869e9125933f42 : SUCCESS in 56m 02s","accounts_in_message":[],"_revision_number":4},{"id":"8e3885f1eb34e00a5f681ffc12ef92071d8b33f4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-10-07 13:08:11.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/07_10_2022_13_01_08\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"bd22da228ef6db113f4fd50df0fb5598e55945fb","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-10-07 13:18:21.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-12-860212-4/1665147659 : SUCCESS in 17m 21s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":4},{"id":"2aed2f2e3dd2629d5b14e2b2c1071a7291ca696a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-10-07 14:02:58.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/12/860212/4/check/DellEMC_PowerFlex/8c48f24/DellEMC_PowerFlex/1406 : SUCCESS in 1h 01m 41s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"a23cd41715288f265a2f94ce2951544a5c151859","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-10-07 14:13:45.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/12/860212/4/upstream-check/cinder-tempest-plugin-ontap-fcp/e13e864/ : SUCCESS in 1h 11m 01s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/12/860212/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/a2af009/ : SUCCESS in 46m 33s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/12/860212/4/upstream-check/cinder-tempest-plugin-ontap-nfs/0be0a0c/ : SUCCESS in 50m 27s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"cc01cdec6d1cdf8f07baf92e0a6198150b1da129","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-10-07 14:18:06.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-12-860212-4/ns5_nfs : SUCCESS in 3h 24m 1s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-12-860212-4/ns5_iscsi : SUCCESS in 3h 26m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"170fe2edd2788434629c42e533200bec8150c8f2","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-10-07 14:22:08.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/12/860212/4/check/DellEMC_SC_FC/f2c35e1/DellEMC_SC_FC/1052 : SUCCESS in 51m 05s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/12/860212/4/check/DellEMC_SC_iSCSI/e63926a/DellEMC_SC_iSCSI/1036 : SUCCESS in 1h 20m 52s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"15f23c6d9a1f87580717468a317a765621c19792","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-10-07 14:35:27.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/12/860212/4/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/af90f2b/ : SUCCESS in 1h 33m 17s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":4},{"id":"32899aad106eaf11cf23940bb6545dc6d90781ff","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-10-07 14:48:42.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/12/860212/4/check/DellEMC_PowerStore_iSCSI/412ebcd/DellEMC_PowerStore_iSCSI/836 : SUCCESS in 1h 27m 30s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/12/860212/4/check/DellEMC_PowerStore_FC/195e3fa/DellEMC_PowerStore_FC/988 : SUCCESS in 1h 02m 05s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/12/860212/4/check/DellEMC_PowerStore_NFS/1acd693/DellEMC_PowerStore_NFS/527 : SUCCESS in 41m 35s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/12/860212/4/check/DellEMC_PowerStore_NVMeTCP/0b16e5a/DellEMC_PowerStore_NVMeTCP/398 : SUCCESS in 1h 47m 23s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"a07c82eef3fca5f92766561e5e3ade3a36ef3d2d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-07 15:28:46.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/12/860212/4/check/DellEMC_Unity_FC/0aedd09/DellEMC_Unity_FC/1915 : SUCCESS in 1h 17m 24s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/12/860212/4/check/DellEMC_Unity_iSCSI/5e2dba9/DellEMC_Unity_iSCSI/1762 : SUCCESS in 2h 03m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"fd663d01cf51931d21efb56c015fe51448c20327","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-10-08 02:45:03.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/12/860212/4/20221008081054/NEC-ISCSI : SUCCESS in 3h 33m 44s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/12/860212/4/20221008081054/NEC-FC : SUCCESS in 3h 34m 06s","accounts_in_message":[],"_revision_number":4},{"id":"3c229b77dcdbfb12dd1c0430e2e48f0c0ec625a5","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-13 12:17:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7cbc47b97952bc00687b1a8aec3877f4a457603a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-10-20 09:44:16.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/12/860212/4/20221020131013/NEC-V-ISCSI : SUCCESS in 2h 42m 42s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/12/860212/4/20221020131013/NEC-V-FC : FAILURE in 5h 34m 01s","accounts_in_message":[],"_revision_number":4},{"id":"e7b45cb06922a1445ac1baf5e29d7ff4aa14b195","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2022-10-21 09:44:02.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4c936e713dae4a414fb289aeb9deeece476423d7","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-10-21 12:27:42.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e81c73e6af424e633fada8f7441b05abacd200ef","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-22 13:16:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"924ea8046d3036aa3629616d5311c7d989ace303","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-25 12:15:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"70b4a10c91bfafed04330a3d67d42bf056d7dbc4","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-27 08:34:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cc440c64e141b2156b17c77eaa1e0128e2c0db0d","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-30 05:37:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5aaf34a2ab3958ad49163a99ce89e37257260764","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-10-30 06:12:00.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"c88e6c15b0d697dd12fa010de754f07f922aa2b3","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-31 13:38:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6c014980098a05f1f31902bb64f71b1d96173bd2","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-11-04 14:39:56.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dca4a775e47ac9a652e8f2ddbdce5ddfbe34e884","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2022-11-10 16:39:34.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3fbdae816f087925b100b4471e868d3232bac614","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-11-16 13:29:49.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"095d8385a492a7498a7ce6a90ef4b4c5ab35e015","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-11-30 14:14:31.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f6e6414a5a85b76bd30d64f634bbdf382351275d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 14:15:15.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"aff5f4e673ed2d6f342e30aaf917f699404cde33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 15:52:14.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/034615d2afb54a34a7f7f0028e90ba80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9594ad39f8c146d2ba5d0b75f7dde3cf : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40a1185ee1c142a5950ff6fe593c992c : SUCCESS in 11m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c529b0080f4a4bdaa87dfc0a9180fdb3 : SUCCESS in 11m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46e00dc392264bac8c112910d276606e : SUCCESS in 12m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/855a60c11ccc4435b4823a25f4dc81a1 : SUCCESS in 1h 02m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b93ecd0b8b054a6c87564365b547cbe6 : SUCCESS in 1h 23m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3febfa9f52aa48fa8ba39199e2ba51dc : SUCCESS in 55m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a4b5c0cb5af044f683aa79c14a7169fc : FAILURE in 51m 38s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/35cf80c4064040948af284a973236d56 : SUCCESS in 1h 26m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9bb77d3d4b9421da080929577f9a043 : SUCCESS in 38m 50s","accounts_in_message":[],"_revision_number":4},{"id":"bdc1f019c0f58d59710152b4b6894cf9bfc78082","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-12-01 03:05:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ec27ab141d5eb77be8ed070e9a0687499ca86f4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 03:52:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f5c25919e69477ebf29c7b45e90a491\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/51f54fba97a649d290a5d47f3f2ad92a : SUCCESS in 21m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b67e8ea49c2c48e7ba29c7e93f06ce42 : SUCCESS in 23m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"03b2719aef43c959845da365cf51375b025b36c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 05:38:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a434e430ec84563a33b4193ef207fb8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18fb0bd2005a42268f3807aa3f5c9f54 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b16a89821f1477fbf96cd254871f7cb : SUCCESS in 10m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4a9dd8149a84558923af206d3897aa3 : SUCCESS in 10m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d696b0b036cb4d60ae685a479d5f91fc : SUCCESS in 13m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/7eef18d83d6d4f59b5d13714490ba4ee : SUCCESS in 48m 13s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/03b3a88a4a4e411ea2e3bbbc8bfc203e : SUCCESS in 50m 45s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a38b75fe5cc645ec9340c0cb8ecbd3a6 : SUCCESS in 1h 25m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a99bb2731ebd45f39b68bf16cd3d353d : SUCCESS in 58m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6a43e2a04c534963b029a41b8882dbd5 : SUCCESS in 15m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2ee3d4f0a9fb418dbb483246dfd3bf43 : SUCCESS in 5m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/10c4b34b293e493497e8911f2647b1b1 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/61aa3ed7ed424a30b7c8bde9b864cbf2 : SUCCESS in 6m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6e4f2530bea8473db7c99e9cd916a837 : SUCCESS in 6m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3d01a2e1ce894bfaa71eaa9917c74e95 : SUCCESS in 51m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e81656df71154d599b883eb8b84cf290 : FAILURE in 6m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4c919627265f4c3ab56e81964e67c2c1 : SUCCESS in 1h 12m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/69eb305381c54743a1032c401c55308e : FAILURE in 1h 23m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/785d50a1398840aea5c286f3948506dc : SUCCESS in 1h 37m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c07749d46bc54fe3b4dfda8812abac09 : SUCCESS in 1h 01m 09s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e0ba3e898c284cd5a95871bdc196e106 : SUCCESS in 1h 22m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/11cd8e7ba38142f18da8714951b518de : SUCCESS in 34m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/48dee0732c60472780bc3c97b114d1da : FAILURE in 18m 35s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/70b193dd7335438fba981dd7e4d52dbc : SUCCESS in 2h 26m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40acad0e8f3144b39071404c8bfc839d : FAILURE in 18m 09s","accounts_in_message":[],"_revision_number":4},{"id":"859873fc0a58dd287c8b678bb57faa197a7a5f0e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-12-01 06:05:07.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/12/860212/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/780b969/ : SUCCESS in 1h 49m 30s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/12/860212/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/48b0714/ : SUCCESS in 1h 49m 45s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/12/860212/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/223d07b/ : FAILURE in 2h 06m 14s","accounts_in_message":[],"_revision_number":4},{"id":"a9f5b38486f99d8b413155d3122792ab5725ead9","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-12-01 06:19:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f16d05e539c0512c2ee9170fec3829702a212279","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 06:49:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4c017400de14c2388929765a49e4ade\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a524278f55034251b3280e47b1560d10 : SUCCESS in 26m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7c525a2287ee4024b1a82a1cb44cc34e : SUCCESS in 25m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ad0e884cbd89a099a01872c5a1a5e605cad380f5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-12-01 08:12:21.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/12/860212/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/c09a03f/ : SUCCESS in 1h 43m 48s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/12/860212/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/bb9f743/ : SUCCESS in 1h 51m 28s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/12/860212/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/a1d8697/ : SUCCESS in 1h 49m 52s","accounts_in_message":[],"_revision_number":4},{"id":"b227231e41e1c2964b0668657313d249484bc694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 08:24:28.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/24071a3dc48e463da7f99a9a8597b775\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/144d53456ffc4656979eba3b02e4db39 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b650a21a35fb4b119817d4b93794b712 : SUCCESS in 11m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ffb2ae48ba184e0783377fda110edc51 : SUCCESS in 10m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54687ddba7c845d680751b81bf37c612 : SUCCESS in 12m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/53a969c8211a4d50b933b9dd609b9833 : SUCCESS in 41m 29s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/70a5b9ac180e42b48c25274b5033d95b : SUCCESS in 53m 19s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ad35667171414b8a80ce8df9bf65e23c : FAILURE in 16m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2893e6f1efe43f1bbf0468da14e56a7 : SUCCESS in 1h 01m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c5da00b73334461680bd57d06792d4aa : SUCCESS in 14m 40s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bfaf94e5bfc9471db9be73c784aae40b : SUCCESS in 5m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6801ebf5ca294049ac65de926a25eee8 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/93ed691af31041d3aa665ab24d83f74f : SUCCESS in 6m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/444c209eeca9491ca68ebf8b1e78fa1b : SUCCESS in 5m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6c59e79662e040889c8336de8fad4dc3 : SUCCESS in 46m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/26576d7cb0a342a28bb796ebbe4528cf : FAILURE in 6m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/86678d24962c43adbf4f09bb8e6c5a30 : SUCCESS in 1h 19m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a36ddf7717964e6e9904400c9afc2054 : FAILURE in 1h 23m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f7c7160c96af41d48a8443fadd8a737a : SUCCESS in 1h 35m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1a9565a39c76480b96c4b03f89d77655 : SUCCESS in 1h 06m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0f70bbd9bf12455bbd0c7fd5469fdf25 : SUCCESS in 1h 43m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ef6afd7d5c0d4296a83d5e9c1a456752 : SUCCESS in 33m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e01e2c9676604747b8b05457cfef92e4 : SUCCESS in 1h 24m 04s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0d4b1a777881425b8856ccff7be2ef89 : SUCCESS in 2h 03m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97a672e5d65849a98149bb766b6da8ad : SUCCESS in 57m 33s","accounts_in_message":[],"_revision_number":4},{"id":"2b9bedab453793f6e87f240a0d15fa630b77d64c","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-12-01 10:26:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a342e0046407bc536352232765c9589368872616","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-12-01 12:23:22.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/12/860212/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/867c5a3/ : SUCCESS in 1h 42m 43s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/12/860212/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/bac82e1/ : SUCCESS in 1h 55m 23s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/12/860212/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/2dd9edb/ : SUCCESS in 1h 38m 59s","accounts_in_message":[],"_revision_number":4},{"id":"313c458fd17ee8b03075b8d9feff4c8252e7826b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 12:41:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9a19c63ef4b47d89ac884a5ae3857a6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0acfa8b87ddc4c87b05b014a511c9375 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3452035d52d40ab9046cfaecd1dd24b : SUCCESS in 9m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9e2a31bc75e46a296395d90e9693139 : SUCCESS in 10m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a21aed1526342f585a60f604f9b7106 : SUCCESS in 11m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/cb72bb8cc866445981b801d805096d7c : SUCCESS in 54m 40s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2fb2cc3ee1a9442496644e9c9076109a : SUCCESS in 50m 29s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2b428fa135fc4ddeb60901f18315897a : SUCCESS in 1h 10m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b70f7811e99c46e4831a068c89c732d7 : SUCCESS in 53m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/178f75b51a564d059974506abd710bd9 : SUCCESS in 15m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6bf22306d5a04d5b86da69c65177c9ce : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2bf7b2b1ea984d4dbde75942e810578b : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/94807bdac31a4417848fbc24ff1de8b7 : SUCCESS in 6m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/62a7bb8bee50496c843e3fce2ae82d89 : SUCCESS in 5m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6db97eefed014bc7900461547247cff3 : SUCCESS in 38m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c85da89b96fc4b1c9a93a1c2ce125d7c : FAILURE in 5m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dbaedda8feaa478f8994b5c2a6e7d5f9 : SUCCESS in 1h 08m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5bc02a306e674535ac1859c4e8e7befd : FAILURE in 1h 28m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eeae78f083ce463ea48b6505fa8b0ee2 : SUCCESS in 1h 31m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3a88f08f67ad4111946f19275d9e14b7 : SUCCESS in 1h 02m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d282fe9f1a5545f4ac15c310547d0983 : SUCCESS in 1h 30m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/52727e92f3c143da9fbfda8d6d5c47fc : SUCCESS in 38m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/366f232224e348cdbb6f125c2dacd598 : SUCCESS in 1h 24m 50s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a5755e59226948cd9f12ec62e802eaa7 : SUCCESS in 2h 01m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/694a8994ee2244fdb3cb4be8dc44a419 : SUCCESS in 42m 09s","accounts_in_message":[],"_revision_number":4},{"id":"c61f7f08437857c258cdb6d48b76be6104bbbc5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 12:42:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"241e0c765ceb220b06d90855c04f335952ebb875","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 13:09:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ac27491cf9c4f268d3ca67cef019821\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/888208923e93476f9588c8f565a87539 : SUCCESS in 31m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bb766f9daaa04fe0a383a7bcc808b2c2 : SUCCESS in 28m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"78e84374cb236dbd6e36e31259bcada0fb964154","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-12-01 13:58:58.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/12/860212/4/check/DellEMC_PowerFlex/6075b20/DellEMC_PowerFlex/1473 : FAILURE in 1h 17m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"2f5efd806a32b06a7937fdc330bc82823f37efae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 14:15:36.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/821f5f3cbe394d588bbb2e38f1d15c89\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbd86fa0e62c48f9b06497011fe80e07 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3faf01bdd7754dbdbb9c9c9fc6f267ae : SUCCESS in 8m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6745fd9ab5e4415db6af556b28d777b5 : SUCCESS in 10m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/216bc6b6e3e64c9387dec15f6c362388 : SUCCESS in 12m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/0c43d9ef62c941ef8b40ecccddaf5ae0 : SUCCESS in 47m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/822a0f71581c47dca9ef7f20fe9f40bc : SUCCESS in 1h 20m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2654b23cda5b4c5e8c3383875bfdb6b8 : SUCCESS in 49m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a552c8fb01614a799bf250f6d62a30a4 : SUCCESS in 1h 03m 38s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/84bfa32132f348f3b6b6d0ce39bfc269 : SUCCESS in 1h 07m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36777cab83d84dd6a45700cb34afcd5b : SUCCESS in 44m 48s","accounts_in_message":[],"_revision_number":4},{"id":"7330d49fcf5492baba3545cf3007db8fc3ddd709","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 14:15:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"905aabca7c3e2662f0ec72383b027015131d17c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 14:18:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34cc58f21a4b427f8f34dfa0f81e82a0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55da74ec19bc435dacd1ca18d1a94c04 : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":4},{"id":"6eaa652a302a2f1b14d9c1b0119f7641556bb1d9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-01 14:35:11.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/12/860212/4/check/DellEMC_Unity_FC/5387c45/DellEMC_Unity_FC/2052 : SUCCESS in 1h 40m 10s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/12/860212/4/check/DellEMC_Unity_iSCSI/1d818fe/DellEMC_Unity_iSCSI/1899 : FAILURE in 1h 53m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6358aa167e2212b62f1250fcfb08e13da53bcb1d","revisions":{"4ce9570b799d925fafb722020059f055471383b8":{"kind":"REWORK","_number":1,"created":"2022-10-04 08:02:35.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/12/860212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/12/860212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/860212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/860212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/860212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/12/860212/1"}}},"commit":{"parents":[{"commit":"c94962b98e599a35e5c00bdbeb699edee19bf33e","subject":"Merge \"setup.cfg: [extras] for pip-installable driver requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c94962b98e599a35e5c00bdbeb699edee19bf33e"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-10-04 07:57:34.000000000","tz":330},"committer":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-10-04 08:02:29.000000000","tz":330},"subject":"update toyou drivers","message":"update toyou drivers\n\ntoyou drivers are currently reporting allocated_capacity_gb.\nThis value is reported by Cinder core.\n\nCloses-Bug: #1991634\nChange-Id: I62c3d41ef92e0d474f5315b5bb1186faa187d958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ce9570b799d925fafb722020059f055471383b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4ce9570b799d925fafb722020059f055471383b8"}]},"branch":"refs/heads/master"},"78c914cd46bc924143ebfaf6fd389c7415ec3265":{"kind":"REWORK","_number":2,"created":"2022-10-04 11:05:40.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/12/860212/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/12/860212/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/860212/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/860212/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/860212/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/12/860212/2"}}},"commit":{"parents":[{"commit":"c94962b98e599a35e5c00bdbeb699edee19bf33e","subject":"Merge \"setup.cfg: [extras] for pip-installable driver requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c94962b98e599a35e5c00bdbeb699edee19bf33e"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-10-04 07:57:34.000000000","tz":330},"committer":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-10-04 11:04:42.000000000","tz":330},"subject":"update toyou drivers","message":"update toyou drivers\n\ntoyou drivers are currently reporting allocated_capacity_gb.\nThis value is reported by Cinder core.\n\nCloses-Bug: #1991634\nChange-Id: I62c3d41ef92e0d474f5315b5bb1186faa187d958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/78c914cd46bc924143ebfaf6fd389c7415ec3265"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/78c914cd46bc924143ebfaf6fd389c7415ec3265"}]},"branch":"refs/heads/master"},"e760cca3e0aac886241d9c2c3ee34f53fc2e4a76":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-10-06 07:20:07.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/12/860212/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/12/860212/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/860212/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/860212/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/860212/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/12/860212/3"}}},"commit":{"parents":[{"commit":"c94962b98e599a35e5c00bdbeb699edee19bf33e","subject":"Merge \"setup.cfg: [extras] for pip-installable driver requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c94962b98e599a35e5c00bdbeb699edee19bf33e"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-10-04 07:57:34.000000000","tz":330},"committer":{"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","date":"2022-10-06 07:20:07.000000000","tz":0},"subject":"update toyou drivers","message":"update toyou drivers\n\ntoyou drivers are currently reporting allocated_capacity_gb.\nDrivers should not report this and let Cinder core take care \nof this value.\n\nCloses-Bug: #1991634\nChange-Id: I62c3d41ef92e0d474f5315b5bb1186faa187d958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e760cca3e0aac886241d9c2c3ee34f53fc2e4a76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e760cca3e0aac886241d9c2c3ee34f53fc2e4a76"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6358aa167e2212b62f1250fcfb08e13da53bcb1d":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-10-07 10:51:07.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/12/860212/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/12/860212/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/860212/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/860212/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/860212/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/12/860212/4"}}},"commit":{"parents":[{"commit":"c94962b98e599a35e5c00bdbeb699edee19bf33e","subject":"Merge \"setup.cfg: [extras] for pip-installable driver requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c94962b98e599a35e5c00bdbeb699edee19bf33e"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-10-04 07:57:34.000000000","tz":330},"committer":{"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","date":"2022-10-07 10:51:07.000000000","tz":0},"subject":"Toyou: Remove allocated_capacity_gb calculation","message":"Toyou: Remove allocated_capacity_gb calculation\n\ntoyou drivers are currently reporting allocated_capacity_gb.\nDrivers should not report this and let Cinder core take care \nof this value.\n\nCloses-Bug: #1991634\nChange-Id: I62c3d41ef92e0d474f5315b5bb1186faa187d958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6358aa167e2212b62f1250fcfb08e13da53bcb1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6358aa167e2212b62f1250fcfb08e13da53bcb1d"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
