)]}'
{"id":"openstack%2Fcinder~862490","triplet_id":"openstack%2Fcinder~master~I10135877889a17c6ae6274f8be1b992294bf65d3","project":"openstack/cinder","branch":"master","topic":"rename_tatlin_api","attention_set":{"29122":{"account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"last_update":"2023-09-01 10:20:24.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"13671":{"account":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"last_update":"2023-09-01 10:10:14.000000000","reason":"Someone else replied on the change"},"31016":{"account":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"last_update":"2023-09-04 07:23:55.000000000","reason":"\u003cGERRIT_ACCOUNT_32029\u003e replied on the change","reason_account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}}},"removed_from_attention_set":{"31779":{"account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"last_update":"2023-07-04 13:00:09.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"30615":{"account":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"last_update":"2023-06-14 05:56:38.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2023-06-14 05:56:38.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"32238":{"account":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"last_update":"2023-07-04 13:00:09.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-06-14 05:56:38.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}}},"hashtags":[],"change_id":"I10135877889a17c6ae6274f8be1b992294bf65d3","subject":"Tatlin unified driver - rename tatlin_api object","status":"NEW","created":"2022-10-24 12:30:59.000000000","updated":"2023-09-10 03:33:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":27,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a1e25c3dfc64a2f6eaa21486eb65954326dfbdda","_number":862490,"virtual_id_number":862490,"owner":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-01 12:17:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":1,"date":"2023-09-04 07:01:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-04 07:23:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-24 12:32:40.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-10-24 12:33:26.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-10-24 12:33:26.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2022-10-24 12:58:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-24 14:00:05.000000000","updated_by":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"reviewer":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"state":"CC"},{"updated":"2022-10-24 14:39:02.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-10-24 14:47:35.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-24 14:49:31.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2022-10-24 15:03:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-24 15:11:18.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-10-24 15:13:51.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-10-24 15:34:54.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-24 15:58:18.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-10-24 16:01:45.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-10-24 16:26:54.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-24 16:27:56.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-24 16:29:01.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-10-24 16:43:13.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-24 16:55:14.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-10-24 17:33:54.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-10-24 18:11:20.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-10-25 02:33:42.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-27 11:59:11.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-10-28 09:31:50.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-01-31 13:41:34.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-31 16:07:47.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"REVIEWER"},{"updated":"2023-01-31 22:08:59.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-01-31 23:05:06.000000000","updated_by":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"reviewer":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"state":"CC"},{"updated":"2023-02-03 03:51:17.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2023-02-03 03:56:18.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-02-15 20:40:47.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-15 23:42:32.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-09 10:01:31.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2023-03-19 06:57:54.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2023-03-23 14:45:57.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-06-14 05:56:38.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"CC"},{"updated":"2023-06-21 04:13:31.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-07-04 13:00:09.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2023-08-29 06:33:22.000000000","updated_by":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"reviewer":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"state":"REVIEWER"},{"updated":"2023-09-04 07:23:55.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2023-09-10 03:33:47.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"}],"messages":[{"id":"310ec063583a90537538f39da70ef202e0cd4f60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-10-24 12:30:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e71f511f0fbcf491fdee465f470978539fa0a9aa","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-10-24 12:32:40.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"76baa7ab6e81cd6c2626b72d01d83b462d0a2237","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 12:58:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9f07eec41b149678a0e01a19d919399\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/606b9149db0f484da9e7db83074e2c2c : SUCCESS in 24m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/17678a55112d4f1f9e6ad226c07b8237 : SUCCESS in 25m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f6e6e0a07d520bce908191117b33ea82d9380010","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-10-24 12:58:51.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0c0945cf1d4d96878067681458d307efd760028f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 13:26:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8714f1253a204a7882b04c49aaff4b0d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a38cc7bcc6954aaf88a76d91212204f7 : SUCCESS in 25m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/edf8b7546cb34799b43b7b29095cbdc0 : SUCCESS in 22m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b61f55b8caafc7ab8df8b1b6cfb2de33526953e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26537,"name":"Cloudbase Cinder SMB3 CI","email":"cinder-smb3-ci@cloudbasesolutions.com","username":"cinder_smb3_ci"},"date":"2022-10-24 14:00:05.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-iscsi http://cloudbase-ci.com/cinder/862490/2/iscsi/ : SUCCESS in 1h 00m 41s\n- cinder-smb http://cloudbase-ci.com/cinder/862490/2/smb/ : SUCCESS in 57m 29s","accounts_in_message":[],"_revision_number":2},{"id":"4d853a12033ddb9fc66a0509b9b2e243985b87b3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-10-24 14:15:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/90/862490/2/check/iscsi-driver-alletra-6k/9965065/9965065.html : FAILURE in 1h 11m 36s","accounts_in_message":[],"_revision_number":2},{"id":"f0f71505a301e11605a5921b0518d903a51c3291","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-10-24 14:39:02.000000000","message":"Patch Set 1:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1666614667874 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"91e089ce277b656dbebb705c48f931815b14b361","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-24 14:47:35.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/862490 : FAILURE in 33m 59s","accounts_in_message":[],"_revision_number":2},{"id":"8b8dc95f5173556e13b042376f180b1d479a06b0","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-10-24 14:49:31.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/8435f5a9311d4b4980463364525d402d : POST_FAILURE in 12m 29s","accounts_in_message":[],"_revision_number":2},{"id":"99e446b628e0f1183bd94041c670fb43c4a986b3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-10-24 14:50:31.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/862490 : FAILURE in 32m 04s","accounts_in_message":[],"_revision_number":2},{"id":"6b58b2752ba70119ea85f6037d665bcfd6fe73a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 15:03:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f77adfb74ac5482b996693bcc054221d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93a774064190456486def9c7109c4b3c : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1b33b60d9594d4dbd1fbb78a4187331 : SUCCESS in 11m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/204cea464de940ca98b5ebf4128c2106 : SUCCESS in 11m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb78f709bdd741509b55e64c218d1ce3 : SUCCESS in 13m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/e5af0200c65444dfb9013a36017eca45 : SUCCESS in 46m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a359ac2dd72c4db69a0a0f447554f3e1 : FAILURE in 53m 10s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f24f8cb1587e483888801aac58b3aab4 : SUCCESS in 1h 20m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16a38eda780a486c8eb3191905329935 : SUCCESS in 46m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a1a87be58c59493982d32b1bf3043fad : SUCCESS in 12m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8afb411525c146d0aad4735028b7c73d : SUCCESS in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f220a836519a474c9ff57a95864abd5c : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/14e2074d1400481c97e89e4751fc8572 : SUCCESS in 6m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/82e3a4ebaa1047b59393c5b19ca7b7b5 : SUCCESS in 6m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b2d6fb71aa664908935a0c5015077793 : SUCCESS in 31m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/afe15286a3794f2ca48b39abd99dc4d6 : FAILURE in 6m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/22fbeadf262940229fd373192d216e11 : SUCCESS in 1h 03m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a26ea5cde8fc401bb028beb2552f4dcd : FAILURE in 1h 24m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9580cfdb48ee45f28a9c3118c5474948 : SUCCESS in 1h 38m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cd6cf892440540ebbc9b7101be5a6b02 : SUCCESS in 57m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dc14cee35a1746eab8be32e67d02e619 : SUCCESS in 1h 31m 07s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3b85ff1bd8334094b071047013d46489 : SUCCESS in 35m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/54af859445844583841d163d992f7b00 : SUCCESS in 1h 32m 44s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2306d99bb10348e5bfd79b87b772064a : SUCCESS in 2h 00m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28ccb7a2d7644122826b6cb9bafad7c8 : SUCCESS in 53m 53s","accounts_in_message":[],"_revision_number":2},{"id":"52fc1b34d3cb7c133717d1e3fca9772a8f6bfcf0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-10-24 15:11:18.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_10_2022_15_03_16\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"2f29bac39cf38b2ef171331831b097a4cbe62ecd","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-10-24 15:13:51.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/862490/2/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/632a499/ : RETRY_LIMIT in 9m 08s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"8fc5bc99c38e8c6906cbce72922fa7bd28475186","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-24 15:34:54.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/4ab7a2018c954a0691875ba51651f5d4 : POST_FAILURE in 2h 09m 45s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/80a541ea5d7f4c6ab2162bf0e4e991c9 : POST_FAILURE in 2h 28m 14s","accounts_in_message":[],"_revision_number":2},{"id":"fac717020d307d75a5615b4a51c5b64f3e77ba1b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-10-24 15:50:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/90/862490/2/check/iscsi-driver-nimble/919e3b7/919e3b7.html : FAILURE in 1h 01m 25s","accounts_in_message":[],"_revision_number":2},{"id":"b100ced2a1d21cae4282f3b55e2c6a86eb7651f0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-10-24 15:58:18.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/862490/2/upstream-check/cinder-tempest-plugin-ontap-fcp/99d078e/ : FAILURE in 52m 48s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/862490/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/b2db960/ : SUCCESS in 39m 27s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/862490/2/upstream-check/cinder-tempest-plugin-ontap-nfs/9cba13f/ : SUCCESS in 42m 20s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"2b79cf8f16da5b07eafa8c6d0ab8103fef14344d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-24 16:01:45.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/90/862490/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/dd32233/ : SUCCESS in 1h 31m 33s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/90/862490/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/9703c4a/ : SUCCESS in 1h 45m 25s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/90/862490/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/386c120/ : FAILURE in 29m 20s","accounts_in_message":[],"_revision_number":2},{"id":"6b0a50a4e17109bac52801304993f2786538df07","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-10-24 16:26:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/90/862490/2/ : SUCCESS in 1h 32m 58s","accounts_in_message":[],"_revision_number":2},{"id":"bfd3751293c51694ae1d87edb2031e87f8f47143","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-10-24 16:27:56.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-862490-1/ns5_nfs : FAILURE in 3h 29m 22s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-862490-1/ns5_iscsi : SUCCESS in 3h 34m 49s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"9ee0e8e605dcc71da3b456e312b43cc308330fc6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-10-24 16:29:01.000000000","message":"Patch Set 2:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/781","accounts_in_message":[],"_revision_number":2},{"id":"04266e61c1527d7068a282095c3a842c713f62f4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-10-24 16:43:13.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/90/862490/2/check/DellEMC_PowerFlex/5e5a1a2/DellEMC_PowerFlex/1436 : SUCCESS in 52m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"6b84d29981830e9a66852f59eceaf5087a05ee19","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-10-24 16:49:50.000000000","message":"Patch Set 2:\n\nBuild Successful \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/ci-result-1666616337996 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"7ec77cbff117baa70b18d17312af2add1ee0ea4e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-24 16:55:14.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/90/862490/2/check/DellEMC_Unity_FC/493f80a/DellEMC_Unity_FC/1975 : SUCCESS in 1h 11m 33s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/90/862490/2/check/DellEMC_Unity_iSCSI/6cc0e3a/DellEMC_Unity_iSCSI/1822 : SUCCESS in 2h 01m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"a52d4f97db6349c2ab01c017cea14622a490f5ae","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-10-24 17:15:33.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-862490-2/ns5_nfs : SUCCESS in 3h 32m 3s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-862490-2/ns5_iscsi : SUCCESS in 3h 37m 13s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"3c04bb7adfb2a62d729de50cee19bf66ed97b104","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-10-24 17:33:54.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/90/862490/2/check/DellEMC_SC_FC/7409e1a/DellEMC_SC_FC/1081 : SUCCESS in 54m 17s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/90/862490/2/check/DellEMC_SC_iSCSI/7359bec/DellEMC_SC_iSCSI/1065 : SUCCESS in 1h 18m 54s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"72da78a9d464a4690862ebe0aa9a810b53449c2a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-10-24 18:11:20.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/90/862490/2/check/DellEMC_PowerStore_iSCSI/97bdbd0/DellEMC_PowerStore_iSCSI/863 : SUCCESS in 1h 24m 32s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/90/862490/2/check/DellEMC_PowerStore_FC/f7088f7/DellEMC_PowerStore_FC/1015 : SUCCESS in 1h 03m 26s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/90/862490/2/check/DellEMC_PowerStore_NFS/d528ed8/DellEMC_PowerStore_NFS/554 : SUCCESS in 41m 15s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/90/862490/2/check/DellEMC_PowerStore_NVMeTCP/24de180/DellEMC_PowerStore_NVMeTCP/425 : SUCCESS in 1h 38m 06s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"cee5243c1b9cd997598cf545445962a5f8b14347","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-10-25 02:33:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/2/20221025044443/NEC-FC : SUCCESS in 3h 32m 28s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/2/20221025044443/NEC-ISCSI : SUCCESS in 6h 48m 56s","accounts_in_message":[],"_revision_number":2},{"id":"b9432b5e156af45d2cf00aabe6b946b4eeb44fd5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-10-27 11:59:11.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/2/20221027144818/NEC-V-ISCSI : SUCCESS in 2h 27m 06s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/2/20221027144818/NEC-V-FC : SUCCESS in 6h 10m 50s","accounts_in_message":[],"_revision_number":2},{"id":"dfe94519300313d0ae7dd8b68e70e829355aa1fa","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-10-28 09:31:50.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 11h 28m 49s","accounts_in_message":[],"_revision_number":2},{"id":"8c1869a6e4dc5c58b072556d016ab7a0d4e8b529","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2022-11-12 18:22:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2af46bdb47a3e81ee0c54d41782b9d0bb1b1c9a2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-11-12 21:30:54.000000000","message":"Patch Set 2:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/813","accounts_in_message":[],"_revision_number":2},{"id":"d2ef0c02b9656fdc5975838ccc481ac2c9a0efc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2023-01-31 12:21:44.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b72619c65118a7d03f548d554bdd5b7bd968d3aa","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-31 12:23:30.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"354ff88495b11798e921ecc3663fb3cdc301d60d","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-01-31 12:28:01.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/e0197ab1f4ad4948ba26459d646a67dc : POST_FAILURE in 5m 37s","accounts_in_message":[],"_revision_number":3},{"id":"f0057991d7a5794697cb868d3ed06a745baf6119","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 12:31:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11e2db11c17044a7b786665966c90b7e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec1b5f0544e1483e8d48f815adf1373f : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/496d94e0fe9545109c0b39627016d480 : SUCCESS in 8m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d340ca750ba8ec5a26094783cf2e0dadfcd9a115","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-31 12:46:55.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/862490 : FAILURE in 23m 26s","accounts_in_message":[],"_revision_number":3},{"id":"88a6a0677d2b4594686f0584857cf9443dd6e339","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-31 12:51:07.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/90/862490/3/check/iscsi-driver-alletra-6k/fcecfd6/fcecfd6.html : FAILURE in 26m 51s","accounts_in_message":[],"_revision_number":3},{"id":"3d2c7b66077bba54b210a1f6054b42efc9746284","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-31 12:51:23.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/862490 : FAILURE in 26m 08s","accounts_in_message":[],"_revision_number":3},{"id":"c9fa9781bd620a80912d961910918a2ca6c602be","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-31 13:41:34.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/116bbe2128e948d5b5a6ed7e85e75bd6 : SUCCESS in 1h 18m 05s","accounts_in_message":[],"_revision_number":3},{"id":"472520fecbf7a316a396f17d7c65a4e6cb803402","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-31 13:43:20.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/90/862490/3/check/DellEMC_Unity_FC/5226ee1/DellEMC_Unity_FC/2179 : SUCCESS in 1h 21m 12s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/90/862490/3/check/DellEMC_Unity_iSCSI/ea6c896/DellEMC_Unity_iSCSI/2045 : FAILURE in 2s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"8f154fdb48abfbed91a2339071a333a6dd611c8d","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-31 13:47:30.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/90/862490/3/ : FAILURE in 1h 08m 45s","accounts_in_message":[],"_revision_number":3},{"id":"70011440e591db1f6192d5af2ff3c48c66c4efd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 14:41:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/becc542135d64148ad4638e57a9da1fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e336fad8bdd949c5b0890af8d9d13d81 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3d43ac5e9f4425caa31c5adfa4d2efc : SUCCESS in 11m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36c400bad5e243998b3634972fdd483c : SUCCESS in 9m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d01b6308495343ac92550d2c39f9779b : SUCCESS in 10m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/5ad93a8909174984b985b83334dbdaa3 : SUCCESS in 49m 03s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/49b0a2f41545473283b3cd9bf73175b7 : SUCCESS in 52m 09s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eac3fbd502e046639233fcda934f9cb9 : SUCCESS in 1h 28m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39bd9800f3e94aa5be9f4bc5d1e59301 : SUCCESS in 1h 23m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/812a03dd0e084fbc920f01c390514b47 : SUCCESS in 17m 46s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ce001f30ada04d2a9ffdc912ac2eee22 : SUCCESS in 4m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/274deb0960b7460b941029cda027a97e : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/92135583d55a4c2cae50db6698d581c1 : SUCCESS in 5m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/803306a2c31e4672b5f8df11e3eb1d49 : SUCCESS in 5m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ae7fa6d8ec414d5781d6b9516234d51d : SUCCESS in 31m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b91958732637402d916ca4ed7c6a5b2d : FAILURE in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/60688c6a9ba2400996adeb3ca5969ca2 : SUCCESS in 59m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4889637f61fe467e84b68a5ae2c54367 : FAILURE in 1h 31m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/db1b37547003427d87dc0893241c8e20 : SUCCESS in 2h 08m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cb5057c7c0714573bbd868ba2baf6d79 : SUCCESS in 1h 03m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/98b4651339a347dfa0f65b4339a3a08c : FAILURE in 1h 31m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/db7968accc53440292c8733fcff8e737 : SUCCESS in 42m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9030d33edb024a2395abf3e202c392eb : SUCCESS in 1h 30m 33s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e9ffe14fc12f4278be7186506e6387ab : SUCCESS in 2h 13m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/342e9f1a0b20453099c93827736a937b : SUCCESS in 45m 25s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0057f8a7bc7748e489d2efbe3b5196cc : SUCCESS in 1h 20m 15s","accounts_in_message":[],"_revision_number":3},{"id":"900d620a4ba8d62498764290dc4b4d4c63c3b4b3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-31 15:38:31.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/cd870c04bdc94683a6290cdcc86ad665 : POST_FAILURE in 2h 12m 06s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/443c3edfcac94e04ac4b953472d2ed23 : POST_FAILURE in 2h 01m 15s","accounts_in_message":[],"_revision_number":3},{"id":"6a6157e49b190aeeb8c659d8253f376fb8d6c4fb","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-31 16:07:16.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/862490/3/upstream-check/cinder-tempest-plugin-ontap-fcp/3514092/ : SUCCESS in 1h 21m 22s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/862490/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/5ee84d5/ : SUCCESS in 42m 07s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/862490/3/upstream-check/cinder-tempest-plugin-ontap-nfs/a745bdb/ : SUCCESS in 45m 07s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/862490/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/66ede2a/ : SUCCESS in 44m 04s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"e7ae29c6155b957fae358b6bc484ed60aaaad007","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-01-31 16:07:47.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b8398e13307437975f14912b6629a7a814da6e45","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-01-31 17:43:25.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/90/862490/3/check/DellEMC_PowerStore_iSCSI/f8627d8/DellEMC_PowerStore_iSCSI/970 : SUCCESS in 1h 18m 03s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/90/862490/3/check/DellEMC_PowerStore_FC/6f3aa04/DellEMC_PowerStore_FC/1105 : SUCCESS in 1h 30m 04s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/90/862490/3/check/DellEMC_PowerStore_NFS/1622449/DellEMC_PowerStore_NFS/609 : SUCCESS in 1h 16m 12s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/90/862490/3/check/DellEMC_PowerStore_NVMeTCP/7e49ad8/DellEMC_PowerStore_NVMeTCP/474 : SUCCESS in 1h 30m 14s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"a704fbfa5bc3e942489a0f92d6f7c62b47727380","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-31 18:09:03.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/90/862490/3/check/DellEMC_Unity_FC/b61ba06/DellEMC_Unity_FC/2183 : SUCCESS in 1h 21m 10s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/90/862490/3/check/DellEMC_Unity_iSCSI/a9e06ca/DellEMC_Unity_iSCSI/2050 : FAILURE in 3s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"e536634cc8b728fa820e68902d00e5d05d30938b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-01-31 20:39:33.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/90/862490/3/check/DellEMC_PowerFlex/e2f9594/DellEMC_PowerFlex/1544 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"c55f00ed75f03c17a6a78ee5007d47aa68a4a182","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-01-31 22:08:59.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/862490-3/HPMSA-iSCSI/ : SUCCESS in 1:37:29\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/862490-3/HPMSA-FC/ : SUCCESS in 1:14:54","accounts_in_message":[],"_revision_number":3},{"id":"8a189c537a3576f35a878d3a288d0e958d8cb9e9","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-01-31 23:05:06.000000000","message":"Patch Set 3:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/90/862490/3/fc : SUCCESS in 2h 51m 32s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/90/862490/3/iscsi : FAILURE in 3h 27m 48s","accounts_in_message":[],"_revision_number":3},{"id":"60ac75560df8aa74106236ba84667d28a67cee70","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-31 23:08:09.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-862490-3/ns5_nfs : FAILURE in 10h 45m 54s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-862490-3/ns5_iscsi : FAILURE in 10h 43m 37s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"fd7ab2cb5f7cfffb9f4d58140486d14447000199","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-02-01 00:30:04.000000000","message":"Patch Set 3:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/1037","accounts_in_message":[],"_revision_number":3},{"id":"528b3395e39aa8ee042d26655c93c66656186d85","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-02-03 03:51:17.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/862490-3/Seagate-iSCSI/ : FAILURE in 1:52:24\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/862490-3/Seagate-FC/ : SUCCESS in 1:31:59","accounts_in_message":[],"_revision_number":3},{"id":"a3f5042f15ed872488b757645443356d7026e9c8","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-02-03 03:56:18.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/862490-3/PVME-iSCSI/ : SUCCESS in 1:52:15\n- PVME-FC https://os-logs.seagate.com/dell-ci/862490-3/PVME-FC/ : SUCCESS in 1:32:03","accounts_in_message":[],"_revision_number":3},{"id":"ec0d95ac3c5ab20aa77bb77f8c7574258aaf6ff4","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-02-10 11:32:22.000000000","message":"Patch Set 3:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/3/20230210173537/NEC-V-FC : FAILURE in 2h 13m 30s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/3/20230210173537/NEC-V-ISCSI : SUCCESS in 2h 56m 41s","accounts_in_message":[],"_revision_number":3},{"id":"e57874abae5e008e4a82255e2ce34fd1ee672523","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-02-14 16:47:42.000000000","message":"Patch Set 3:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/14_02_2023_16_39_34\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"e601a4bb99f2c4006d4f1e4b39747d40bbfd1523","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-02-14 16:56:30.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/14_02_2023_16_48_39\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"d216d87283463a7a145f29173a9c057e47327fad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2023-02-15 15:21:31.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"e1207052684a23a9a7b531e684a10d38c38f7e8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2023-02-15 15:21:52.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"0ae8682030aa76a68524c07f8e319f33d24945d1","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2023-02-15 15:35:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"736608c540ba2b3923dd1ac236e52d5fed52bdb7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-15 15:38:00.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"079cb608d62a663a4ae3f295aa7e24cc96dd2e23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 15:39:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc805adebcbd4248bc60013a0f78003e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed6540283aaf499bab742362d46aad60 : SUCCESS in 16m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/302630e7d8b74b7c9f7d83d9c67745f6 : SUCCESS in 10m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"35c510345c712cef9f2a7b3160d279a2db8227ae","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-15 16:38:13.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/862490 : FAILURE in 23m 23s","accounts_in_message":[],"_revision_number":5},{"id":"67194fdabc39b5c6f8e6e4b732072729d6d55c03","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-15 18:06:52.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/90/862490/5/ : FAILURE in 1h 04m 37s","accounts_in_message":[],"_revision_number":5},{"id":"3ad9d6ea64588e03d43e1b4732a96ade5f144093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 18:19:16.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/93a05869dc5f432cbe18e4214c3fda8c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49e3f8ab28a3492abe49eb00dd79c2d6 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef8d41b7292e4f0282aae3eb631c63ab : SUCCESS in 11m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d91a8364af54d3e8efac1550a6f23a3 : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/829f7778917943d8a672487bbb1f2853 : SUCCESS in 14m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/de1e9cc075ef4f6fa9a398f4c3a83b89 : FAILURE in 50m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/cac72f24c52847c5912902207fe54058 : SUCCESS in 1h 05m 07s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/177c42de1d1f41de9908155e08791c23 : SUCCESS in 1h 22m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29308459abdf4b848d058531f387fabd : SUCCESS in 1h 06m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ce7e2f18c34749dc9c4fe0e17dc3c2a1 : SUCCESS in 15m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2d52211144b94301b49feda8371849a6 : SUCCESS in 5m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5698a334de364e4e919a604f13fed4a8 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bd4a524ce6614e6aabe5b210b3f1136b : SUCCESS in 5m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5fa8644dfd424c51b586c6a9a382772b : SUCCESS in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6d0d3825ee5546c7a1a739a453c2e70f : SUCCESS in 57m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/42eae3e927d148959208154bff7c369c : FAILURE in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bea7c583cb9f4883869b25df047e9fd5 : SUCCESS in 2h 00m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b05190c963d849869d8d89f47edc758a : FAILURE in 1h 41m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/085b316f30d041ebacd7228250c8b943 : SUCCESS in 2h 43m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/01137a057a994868b4ac10e0563b9e48 : FAILURE in 1h 33m 24s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8da99c6be58e4af19afea447954eb4bf : SUCCESS in 1h 23m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/61afad746276497b806271b134ea3371 : SUCCESS in 35m 15s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a9b522cd563940a2b874629d8e16f43d : SUCCESS in 1h 18m 22s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d056c11b429d4b3d8ccdb6a78ecb1047 : SUCCESS in 1h 50m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc1027e5f2304833adbfc666f8a7e3a6 : SUCCESS in 54m 17s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed8511bfddb742478a576de044693a09 : SUCCESS in 1h 12m 02s","accounts_in_message":[],"_revision_number":5},{"id":"80e7106b0a748b20e5e8cad174a5c38adf5327db","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-15 18:36:08.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-862490-4/ns5_nfs : FAILURE in 1h 58m 52s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-862490-4/ns5_iscsi : FAILURE in 1h 58m 53s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"a5344f5d6a6ad38159730108183ee5dd04a84b55","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-15 18:53:39.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-862490-5/ns5_nfs : FAILURE in 1h 53m 50s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-862490-5/ns5_iscsi : FAILURE in 1h 53m 47s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"f1cb62707d26474719dc10604dcc259ae05ecf61","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2023-02-15 20:40:47.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/90/862490/5/check/dsvm-tempest-zadara-driver/8d2f63e : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 20m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"aec1d811060b31871429529409552a77781b726e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-15 21:01:03.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/90/862490/5/check/iscsi-driver-alletra-6k/078e45a/078e45a.html : FAILURE in 26m 02s","accounts_in_message":[],"_revision_number":5},{"id":"046cd381cce922b74d92c023f76429030a0b8c98","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-15 21:50:38.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/862490 : FAILURE in 26m 26s","accounts_in_message":[],"_revision_number":5},{"id":"dd24e5e55c4859575271ff2db422badc0ed778df","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-02-15 23:42:32.000000000","message":"Patch Set 5:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/90/862490/5/check/fujitsu-eternus-dx-iscsi/2991979 : FAILURE in 2h 47m 40s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/90/862490/5/check/fujitsu-eternus-dx-fc/a72c34f : SUCCESS in 1h 51m 21s","accounts_in_message":[],"_revision_number":5},{"id":"6afd360a7b04b02971ad31878769b1880050b7a6","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-16 01:01:01.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/21a7caa9abdc4283b9b2f1ed17c4e518 : SUCCESS in 1h 07m 05s","accounts_in_message":[],"_revision_number":5},{"id":"0762a4d15657b8a544a0ee8ff7808d3628fe1108","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-16 07:13:38.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/90/862490/5/check/DellEMC_Unity_FC/a5f28e0/DellEMC_Unity_FC/2291 : SUCCESS in 1h 26m 13s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/90/862490/5/check/DellEMC_Unity_iSCSI/a09b64e/DellEMC_Unity_iSCSI/2150 : SUCCESS in 1h 48m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"9942cfcabbdd0818db50fe2f13605455577bb50e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-18 01:47:38.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/622f1c30b75243ca8255d994a2206966 : RETRY_LIMIT in 1m 01s","accounts_in_message":[],"_revision_number":5},{"id":"e0e52d0f85250ab2c0cca6a0de8a06a3715f4678","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-21 13:54:11.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7a7a875a325343fe9c5f24c29f844bd8 : SUCCESS in 1h 16m 24s","accounts_in_message":[],"_revision_number":5},{"id":"7f0e1c5984a96ad1efa67929d354820b3a5c5a34","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-02-21 23:56:00.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/5/20230222062745/NEC-V-ISCSI : SUCCESS in 2h 26m 56s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/5/20230222062745/NEC-V-FC : SUCCESS in 2h 28m 11s","accounts_in_message":[],"_revision_number":5},{"id":"5e2dfa5b3e7ba2dc5daa8e125917906ce482e4a5","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-02-27 01:08:54.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/5/20230227063128/NEC-FC : SUCCESS in 3h 31m 56s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/5/20230227063128/NEC-ISCSI : SUCCESS in 3h 37m 23s","accounts_in_message":[],"_revision_number":5},{"id":"d67fdf480db215c28ef168af0bf8e72958562817","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2023-03-01 16:31:32.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"f805333fce0a5f3b43c72e916732d67330c160cb","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2023-03-01 16:33:46.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f91d3e04444d2a5372578daf10675b0583cd5344","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 16:53:28.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bc160b068f04962aa3727ba86446ad1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/834812bde72b4dc183273106d0062ff7 : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4b49834b8bfd4ead9aa7b02981d6a91f : SUCCESS in 18m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"23de53ad4e9d5cbfa8d3b922ad23af9d3312e06d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-01 17:23:08.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/862490 : FAILURE in 23m 38s","accounts_in_message":[],"_revision_number":6},{"id":"ddfa069ce2987c4339647b12a029db6f04bf2a98","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-01 17:27:31.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/90/862490/6/check/iscsi-driver-alletra-6k/c3d3f61/c3d3f61.html : FAILURE in 26m 05s","accounts_in_message":[],"_revision_number":6},{"id":"9f0d1c72f57b20437dc9ba28b98dd18760ae9596","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-01 17:31:47.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/862490 : FAILURE in 26m 27s","accounts_in_message":[],"_revision_number":6},{"id":"9d77969c8eb98e442209835c2ebdd33b07a5d836","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-01 20:18:28.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/94a7bd18bc344bb3811a9d894c7cf19a : SUCCESS in 1h 10m 50s","accounts_in_message":[],"_revision_number":6},{"id":"164c56a5664a2c90a0e8c02917e82d67c63b7e8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 22:11:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c9d1049663a4fd2a3f50cac540c53cb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a380aac5f4fc4ddaa115928193531978 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba34ae22957b406fbf8890a0c07ba26e : SUCCESS in 10m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8dd97b2b9afd4c0da276212492cb555a : SUCCESS in 11m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2aa57f78d6c4665a399b921f94d067d : SUCCESS in 12m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/a6c55fc31b864b029c4b1cfb190ff91b : SUCCESS in 57m 45s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e39609fc5b64409e826c69bc16049cb4 : SUCCESS in 57m 24s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2ad97d3843f144e78141f35dbd1d8a70 : SUCCESS in 1h 26m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4dd67887ccc94210a4a3ed5ce427eac9 : SUCCESS in 1h 00m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/61c386ee5dbe4e018868a63df3fbee60 : SUCCESS in 16m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/724040ca8b39403d8f45f048639550e1 : SUCCESS in 4m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d9c31d116d8849d893c53ca18348695e : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/43de2809ac4346e493c92b4d680319d2 : SUCCESS in 5m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b981996b4f774921a4321539b836af4d : SUCCESS in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/807e26f50fba464b86bf854f9672475e : SUCCESS in 39m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4a36e876dc12404abe957b01c8f33414 : FAILURE in 6m 17s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/71744d145d1e45a694ec81c9f996ce5e : SUCCESS in 1h 17m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/347ce594c1ed4c94a6061e76747058a1 : FAILURE in 1h 27m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b5c4f75323a248bbba9c9143c4283aa9 : SUCCESS in 1h 46m 42s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b72d54db3f3744bcb4dc8039547ce0ed : SUCCESS in 1h 10m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/863cf36916ef4cd193c1cb14463d8d3c : FAILURE in 1h 26m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/819de1f3cb464fec8b76a41cd878ebef : SUCCESS in 36m 48s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9096c47d72bb4528b105b4c4324819d4 : SUCCESS in 1h 18m 58s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/10868ede10d043a98e3e6874b3253ad5 : SUCCESS in 2h 54m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c05c33ce4e546249f044a9713049b3a : SUCCESS in 1h 34m 53s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10d2b04d61ae45608cfac6677e212800 : SUCCESS in 1h 35m 20s","accounts_in_message":[],"_revision_number":6},{"id":"c5a02cb6982f078b6086909a9d19750bd1273f4d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-03-01 22:19:49.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/01_03_2023_22_11_49\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"783d2f6cd99d617da0bd45d17f1c05a0ede56166","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-01 23:28:05.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-862490-6/ns5_nfs : FAILURE in 4h 23m 31s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-862490-6/ns5_iscsi : FAILURE in 4h 54m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"661eb63748f006c37a9fb0ff008db9459e4941f7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-03-01 23:42:12.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/862490/6/upstream-check/cinder-tempest-plugin-ontap-fcp/f6806a6/ : SUCCESS in 1h 29m 34s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/862490/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/04c75f1/ : SUCCESS in 44m 45s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/862490/6/upstream-check/cinder-tempest-plugin-ontap-nfs/32eed10/ : SUCCESS in 42m 00s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/862490/6/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/cc58d30/ : SUCCESS in 1h 16m 28s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"95f97e2fd63f97ca95fe6b51173fc35a39dec184","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-02 00:55:13.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/c8760e630b7d426fa898d9f42a3ef7cc : SUCCESS in 2h 10m 39s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"b6970c1c34124b20f68d6bb36f34c4ca7873aba1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-02 05:55:41.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/90/862490/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/6733fbb/ : SUCCESS in 1h 45m 13s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/90/862490/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/2c1b801/ : FAILURE in 2h 02m 05s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/90/862490/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/5de5ab9/ : SUCCESS in 1h 54m 16s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/90/862490/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/7bfeecf/ : SUCCESS in 1h 42m 17s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b424d98bf492f8bc65506cbf67b030efa7cdde62","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-03-02 10:51:08.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/90/862490/6/ : FAILURE in 1h 09m 26s","accounts_in_message":[],"_revision_number":6},{"id":"b69e6091bf136062c85edf915aa3dea0e1b7cfc3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-03-02 14:57:56.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/90/862490/6/check/DellEMC_PowerFlex/fe73985/DellEMC_PowerFlex/1685 : SUCCESS in 1h 40m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"411614d64b5a1008f2d77cf57418b43558e2d4e7","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-03-02 15:22:44.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/90/862490/6/check/DellEMC_PowerStore_iSCSI/f6c010c/DellEMC_PowerStore_iSCSI/1079 : SUCCESS in 1h 13m 32s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/90/862490/6/check/DellEMC_PowerStore_FC/ba1043e/DellEMC_PowerStore_FC/1202 : SUCCESS in 1h 32m 05s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/90/862490/6/check/DellEMC_PowerStore_NFS/04e1af4/DellEMC_PowerStore_NFS/718 : SUCCESS in 1h 16m 17s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/90/862490/6/check/DellEMC_PowerStore_NVMeTCP/df2db03/DellEMC_PowerStore_NVMeTCP/583 : SUCCESS in 1h 29m 12s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"e8e4c89edfc081f57cb30a2b3db32519f948712f","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2023-03-02 16:38:24.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/90/862490/6/check/DellEMC_SC_FC/eef785e/DellEMC_SC_FC/1150 : SUCCESS in 1h 04m 16s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/90/862490/6/check/DellEMC_SC_iSCSI/2b8a102/DellEMC_SC_iSCSI/1134 : SUCCESS in 1h 29m 38s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":6},{"id":"18d180114085c491b3899ad338734ea924a5c786","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-02 23:30:39.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/90/862490/6/check/DellEMC_Unity_FC/b85c658/DellEMC_Unity_FC/2407 : SUCCESS in 1h 30m 36s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/90/862490/6/check/DellEMC_Unity_iSCSI/9502b44/DellEMC_Unity_iSCSI/2257 : SUCCESS in 1h 59m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"3a3752748a244fadd8b41fbdaabc45aff9f141bf","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-03-04 14:20:39.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/862490-6/HPMSA-iSCSI/ : SUCCESS in 1:36:31\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/862490-6/HPMSA-FC/ : SUCCESS in 1:17:22","accounts_in_message":[],"_revision_number":6},{"id":"38b69a58b320dbef72806b322aeeadd032255fe0","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-03-04 22:07:14.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/862490-6/PVME-iSCSI/ : FAILURE in 1:54:19\n- PVME-FC https://os-logs.seagate.com/dell-ci/862490-6/PVME-FC/ : SUCCESS in 1:36:08","accounts_in_message":[],"_revision_number":6},{"id":"a3733cc3e8e2768d25d050ecc7ddad7fa2e3b9c9","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-03-05 02:12:22.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/862490-6/Seagate-iSCSI/ : SUCCESS in 1:53:02\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/862490-6/Seagate-FC/ : SUCCESS in 1:34:46","accounts_in_message":[],"_revision_number":6},{"id":"94104b4cf112c965bd2582e075e2c5afdec40701","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-03-07 15:40:57.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/6/20230307213548/NEC-V-ISCSI : SUCCESS in 2h 42m 55s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/6/20230307213548/NEC-V-FC : SUCCESS in 3h 05m 06s","accounts_in_message":[],"_revision_number":6},{"id":"586d698d924c955698a602ba33a98f5bcb9abc4f","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-03-09 10:01:31.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"43ad8aa768f841cb5daea5307542d6cd069b59b1","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-03-09 23:58:15.000000000","message":"Patch Set 6:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/90/862490/6/fc : FAILURE in 1h 13m 33s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/90/862490/6/iscsi : FAILURE in 0h 42m 06s","accounts_in_message":[],"_revision_number":6},{"id":"9c968cd2b2ea5ab721aec570dbb28d0ee85fdb05","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2023-03-10 12:03:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"95233473487141c8ac81783666ded78d3ddfabf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2023-03-10 12:08:10.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"54bf7bc902d53992859d407c8b93f828f0f5169c","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-03-10 12:08:34.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"3303a2efeaccba0c4fc7d3ea6dc10ad367d0eae6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-10 12:08:48.000000000","message":"Patch Set 7:\n\nBuild Started http://openstack-ci.spb.yadro.com:8080/job/openstack-gerrit-trigger-fc/78/ (1/2)","accounts_in_message":[],"_revision_number":7},{"id":"d8709c2ad041e3df454ea404de04b2cdec4540f2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-10 12:29:09.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder finger://openstack-ci-fc/11c7426485b6471a92836a8071fcb69e : RETRY_LIMIT in 46s","accounts_in_message":[],"_revision_number":7},{"id":"2f02468839d402801569db5ca848708f07be22a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 12:30:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1f78b0ee566434190428ac22f0f671e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/42ed16ee343e456dba5d3b3a518f8715 : SUCCESS in 20m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fca09c15331e436fa280b1cbcccba92a : SUCCESS in 19m 43s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ce7ac6ead5c434d4c69408b3a5fe7351dca37a1c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-10 12:37:07.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/862490 : FAILURE in 27m 12s","accounts_in_message":[],"_revision_number":7},{"id":"050e901ae7f8e40419163e668b95e1acf1ea628c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-10 13:22:22.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ea2855ec445f43e1bccbfb3a8e4ba089 : SUCCESS in 1h 12m 36s","accounts_in_message":[],"_revision_number":7},{"id":"419f763a42678d156a312afa9396219e56404796","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-10 14:11:54.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/90/862490/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5c89998/ : SUCCESS in 1h 44m 46s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/90/862490/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6223e81/ : FAILURE in 2h 00m 43s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/90/862490/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/d38be4d/ : SUCCESS in 1h 41m 31s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/90/862490/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/ac44201/ : SUCCESS in 1h 58m 08s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2938c72d44c7d2d0903b5d8ab6d4f2cd0738fa2d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-10 14:17:37.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/90/862490/7/check/iscsi-driver-alletra-6k/b0b99d2/b0b99d2.html : FAILURE in 1h 57m 40s","accounts_in_message":[],"_revision_number":7},{"id":"ebb008e35b19b8d3233ca2364d44b866398e95f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 14:23:40.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83640b9978484f08bb8c969e2d64d578\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/befe7db07451445c8362d4c49f9f8151 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66cee5a23b1e4147ae96afc135d848b7 : SUCCESS in 10m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebb7fb4ca5c34c6facf95086b7e31aaf : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5e3d6dddbb34dc69ff24ade88bf643a : SUCCESS in 12m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/143c399d498f4b6f91af2d24b26cb41e : SUCCESS in 53m 40s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/cbe84820a9d04efcb1c7fa70f419a848 : SUCCESS in 53m 54s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dc8645013eb94f10beca18f1c1d812a0 : SUCCESS in 1h 32m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7fe33123c0247489a3e4e7f4993d470 : SUCCESS in 52m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e36ada56cd5f4ea6bca62b0d0278807c : SUCCESS in 13m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8ec28e25b77d4ce6bb43a4e4071bfb97 : SUCCESS in 4m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a2b1d4a496f54f6581e3f77107805e5e : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/942a9f95f112464ea8bebb764d37c85a : SUCCESS in 5m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9c7cf836e05c45759adb258089ab6881 : SUCCESS in 5m 44s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/753c82812b3549ebba368bbad832ea5c : SUCCESS in 57m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2b184a090744434486a509d5bd8ecc41 : FAILURE in 5m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a1500dbfa3e74f13a72e743b6e0210c0 : SUCCESS in 1h 11m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f21a2cb564cf46db96d3580283a9c03b : FAILURE in 1h 25m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/11564bdc10984ae7ae190a5994f6d7bb : SUCCESS in 2h 09m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7343df181eff4931a05012c53876ae94 : SUCCESS in 1h 02m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/360254292e864b25a913cca0b9ac68c5 : SUCCESS in 1h 35m 52s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d492d10926ef46ee8c70a10b9fe4687b : SUCCESS in 28m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3ba1d0d078164ce6b18b07682262beee : SUCCESS in 1h 54m 11s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fb8107fdef2e4f0b84ee63d73adb5174 : SUCCESS in 2h 03m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff8d385e9fcb4b7083c3ef61eff60566 : SUCCESS in 1h 02m 13s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e012cf10b81144f7b4231fc2e7381e40 : SUCCESS in 1h 21m 32s","accounts_in_message":[],"_revision_number":7},{"id":"f3a7f5c10ddcc2494899b24bf0386649a880dfb0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-03-10 14:31:51.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/10_03_2023_14_23_50\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"a925a67516c6d13abab5168e89dd3c9648051759","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-10 14:41:51.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/9dfcabe2b4bf4c48a01b72d7602741a2 : SUCCESS in 1h 16m 04s","accounts_in_message":[],"_revision_number":7},{"id":"4fb47330660d4978b99051a40db19c3b278ca829","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-03-10 15:27:19.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/862490/7/upstream-check/cinder-tempest-plugin-ontap-fcp/14803d2/ : SUCCESS in 58m 01s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/862490/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/52d9176/ : SUCCESS in 1h 02m 03s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/862490/7/upstream-check/cinder-tempest-plugin-ontap-nfs/3ef0f82/ : SUCCESS in 44m 30s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/862490/7/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/3fe2154/ : SUCCESS in 45m 17s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"5885f99885117f9fbebc69f1a184b05a2f176a46","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-03-10 15:47:35.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/90/862490/7/ : FAILURE in 1h 06m 46s","accounts_in_message":[],"_revision_number":7},{"id":"6ae7722986295e2f56a62013671e745c8644e3d0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-03-10 15:57:49.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/90/862490/7/check/DellEMC_PowerFlex/67967f0/DellEMC_PowerFlex/1731 : SUCCESS in 1h 33m 51s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"96ee269e22cd4a7a8e05ba73f482d9b2da0c9b5c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-03-10 16:04:21.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/90/862490/7/check/DellEMC_PowerStore_iSCSI/8f008c2/DellEMC_PowerStore_iSCSI/1125 : SUCCESS in 1h 15m 44s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/90/862490/7/check/DellEMC_PowerStore_FC/5165bca/DellEMC_PowerStore_FC/1247 : SUCCESS in 1h 31m 09s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/90/862490/7/check/DellEMC_PowerStore_NFS/89d34fc/DellEMC_PowerStore_NFS/764 : SUCCESS in 1h 21m 14s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/90/862490/7/check/DellEMC_PowerStore_NVMeTCP/16fa4e6/DellEMC_PowerStore_NVMeTCP/628 : SUCCESS in 1h 32m 36s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"fa61548aebcc5eb7bbb67198b3f5a538c3695c0d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-10 16:21:01.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/00b5ad29ab64467bada06e0ea0b74b42 : SUCCESS in 1h 59m 15s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/3f738c8dd35a4e3497efccc1d0eba819 : SUCCESS in 1h 39m 02s","accounts_in_message":[],"_revision_number":7},{"id":"8114457429857a5c995dc56f9a96cc6266dccd5e","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2023-03-10 16:41:31.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/90/862490/7/check/DellEMC_SC_FC/d0c9384/DellEMC_SC_FC/1198 : SUCCESS in 1h 02m 17s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/90/862490/7/check/DellEMC_SC_iSCSI/41b5060/DellEMC_SC_iSCSI/1178 : SUCCESS in 1h 44m 39s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":7},{"id":"a4f5556a96a8cf0b1189b53d1261a020422adf78","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-10 17:03:19.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-862490-7/ns5_nfs : FAILURE in 4h 15m 30s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-862490-7/ns5_iscsi : FAILURE in 4h 37m 59s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"b6ff4defe40405311ef20f27b63d789e08d44303","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-10 19:30:50.000000000","message":"Patch Set 7:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/78","accounts_in_message":[],"_revision_number":7},{"id":"26c9f9f7ea4581b3928576720e0c39bab1791b84","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-10 20:26:12.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/90/862490/7/check/DellEMC_Unity_FC/b8b523e/DellEMC_Unity_FC/2475 : SUCCESS in 1h 36m 30s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/90/862490/7/check/DellEMC_Unity_iSCSI/8ba8b95/DellEMC_Unity_iSCSI/2324 : FAILURE in 1h 18m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"b5a3448c4b4bc7fcd366988a5b730b4ebb9323b9","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-03-10 22:25:29.000000000","message":"Patch Set 7:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/90/862490/7/fc : FAILURE in 0h 35m 29s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/90/862490/7/iscsi : FAILURE in 0h 42m 01s","accounts_in_message":[],"_revision_number":7},{"id":"9705154b4cb32db572b708ea7762c01b2913a062","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-03-12 03:31:38.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/862490-7/HPMSA-iSCSI/ : SUCCESS in 1:36:21\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/862490-7/HPMSA-FC/ : FAILURE in 1:15:56","accounts_in_message":[],"_revision_number":7},{"id":"199dfa43c922062a4913a04fb1e51c9cbb360bab","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-03-12 23:49:24.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/862490-7/PVME-iSCSI/ : SUCCESS in 1:45:05\n- PVME-FC https://os-logs.seagate.com/dell-ci/862490-7/PVME-FC/ : SUCCESS in 1:32:42","accounts_in_message":[],"_revision_number":7},{"id":"aaf8e8f0e9497ea5409f9ce67d2e0bfc51bb5781","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-03-13 11:08:35.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/862490-7/Seagate-iSCSI/ : SUCCESS in 1:44:49\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/862490-7/Seagate-FC/ : SUCCESS in 1:33:03","accounts_in_message":[],"_revision_number":7},{"id":"cf2897953a33bd08ac80a60ae5f7feeef088e6c2","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-03-16 13:05:16.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/7/20230316153433/NEC-V-ISCSI : SUCCESS in 2h 26m 40s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/7/20230316153433/NEC-V-FC : SUCCESS in 6h 30m 39s","accounts_in_message":[],"_revision_number":7},{"id":"8a3d3676bff0c01da53a33c02e5aca4eef93aa08","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-03-19 00:51:23.000000000","message":"Patch Set 7:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/7/20230319023528/NEC-ISCSI : FAILURE in 7h 02m 58s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/7/20230319023528/NEC-FC : FAILURE in 7h 15m 52s","accounts_in_message":[],"_revision_number":7},{"id":"7acef4cccca1ccc16ba659ce5eb0e97e6056af47","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-03-19 06:57:54.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"0c54df1203995fc96be6b4bbdf3001bbe1610ecf","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-03-22 08:04:47.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"dc746adde58a1dadcea2235815efec8467d354e8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-03-23 14:45:57.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-90-862490-7/FC/ : SUCCESS in 2h 38m 24s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-90-862490-7/iSCSI/ : SUCCESS in 2h 15m 39s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"c30df9ad170903b239f3baf8313ea6923a6af650","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-21 22:08:18.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/90/862490/7/check/iscsi-driver-nimble/7a172d6/7a172d6.html : FAILURE in 1h 48m 45s","accounts_in_message":[],"_revision_number":7},{"id":"ded0f858f59124b50551a918c181b13441820e86","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2023-05-22 12:04:20.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"99136fa092407cf65388a72bdbc2d4127569e45f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-05-22 14:42:58.000000000","message":"Patch Set 7:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1360/","accounts_in_message":[],"_revision_number":7},{"id":"0fe6e1b37592d8d93df245b984d28fd3377e484d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2023-06-13 13:52:00.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"99211ca5604e03efc917646b0311a19c3bbe3646","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-13 13:52:55.000000000","message":"Patch Set 8:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":8},{"id":"14b5406f020c2ede3c43be2843ca58985b4d11b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 14:08:32.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e90de9228e0f4628b5b30794505a377e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/154f5dcb5c714327bedbfe2b6c820f58 : SUCCESS in 15m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b5a6cbc24e448d7ad8d98325cf9d8ad : SUCCESS in 10m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/57523d4645414c339d9ea80e2aa28e79 : SUCCESS in 8m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8e91760a4bc7a2cc32b96d93da6a6e5ff6e64a31","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-13 15:08:08.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/90/862490/8/check/iscsi-driver-alletra-6k/4ddea8b/4ddea8b.html : FAILURE in 59m 23s","accounts_in_message":[],"_revision_number":8},{"id":"23d6dd5021f1a9215aa397ca9540866ec285e4b2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-13 15:09:55.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/0d5fa3978d91449ca1479316459b41dd\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/d3fa33e267dd4ee7bbd4598af72d874e : FAILURE in 1h 15m 12s","accounts_in_message":[],"_revision_number":8},{"id":"769df57d4369fbfe5f2d8cb6987992eab7f8b4f0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-13 15:21:03.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/862490 : SUCCESS in 1h 25m 02s","accounts_in_message":[],"_revision_number":8},{"id":"3e3f09374deec103f9ece642b22e8d6e4ae17743","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-13 15:45:36.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/90/862490/8/check/DellEMC_Unity_FC/45ee537/DellEMC_Unity_FC/265 : SUCCESS in 1h 52m 41s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/90/862490/8/check/DellEMC_Unity_iSCSI/a4a70d0/DellEMC_Unity_iSCSI/265 : SUCCESS in 1h 53m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"da694b7f76e4c094d2e042cb9b57d15a3c0d76c8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-13 16:39:16.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/90/862490/8/check/iscsi-driver-nimble/e060b9e/e060b9e.html : FAILURE in 1h 05m 57s","accounts_in_message":[],"_revision_number":8},{"id":"1faf1502ef611da986713b962e2a1b6a0579ca14","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-13 16:48:07.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-90-862490-8/FC/ : SUCCESS in 2h 35m 15s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-90-862490-8/iSCSI/ : SUCCESS in 2h 41m 40s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"dfa8700153506ac512d559dee41ecf0ba4366a2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 17:02:55.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/34fb953132f048088ee7694c25d4a4d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c8f1f15f99f4dab9422915b6747701c : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b7402b8e64749f0a723ca79cb736b6a : SUCCESS in 12m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/018764aa744f4247bc2aff9526b53c0a : SUCCESS in 11m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c5da7714ce84a0fae554d6d7482f5dc : SUCCESS in 11m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa773f11cc684cde9397179b4588c707 : SUCCESS in 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/1970ac5df962409fa36fed02a33842a9 : SUCCESS in 42m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2625ad54c0774d219ea99a041826c42d : FAILURE in 1h 27m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80db6e680a844c1cb52b27040758955c : SUCCESS in 49m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8ed3c1acf00749b495154f7a040ee465 : SUCCESS in 16m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/65f1519750e849188cb0e56787f98490 : SUCCESS in 5m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/87743394fb564bfd9dde7546603fdd78 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4d64d6f621a141d284b35144f16db2fd : SUCCESS in 5m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c8145ca6dde848098bb5bbd77af2cd07 : SUCCESS in 6m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1034981ebbf24ba085ed72d90aa09509 : SUCCESS in 38m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b982338e2d0247ce9ed47cc744c7219b : SUCCESS in 5m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/90a77f037f004b69a06e6ba99f0f9ce7 : SUCCESS in 1h 52m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5d90f304bc35458aa807f341ad8157d6 : FAILURE in 1h 47m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/58dad2483b58432aa3355c30c5208e27 : SUCCESS in 2h 01m 15s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/d39502601bc8459e8ac0fd62102fd6a6 : SUCCESS in 2h 20m 56s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e1abaebef4ba491ba64b4da58b2cb234 : SUCCESS in 57m 58s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/96db98edc1404961a9f46f51141f234d : SUCCESS in 1h 54m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5b08ab866464450aa512b98fc5f91981 : SUCCESS in 39m 55s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6a50a7e326774059bf8b20691e5c20ac : SUCCESS in 1h 33m 19s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/86b4ff54f0b0455ca35579aa9ada353e : SUCCESS in 2h 14m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b8257a33601a43b2a22ca3fed1d84d62 : TIMED_OUT in 3h 03m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61c1ddac8e18413993d6f127938b4178 : SUCCESS in 1h 07m 31s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/6914d74b79a54e01a5241e6a05c7d8d5 : SUCCESS in 12m 46s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c7cc5dd644228c4f5f3aa02e244b410a95e7c829","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-13 17:58:20.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-862490-8/ns5_nfs : SUCCESS in 4h 5m 47s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-862490-8/ns5_iscsi : SUCCESS in 3h 58m 6s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"00addb36bb8a10c5acda970c2f5d64fdedbd2276","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-06-14 05:56:38.000000000","message":"Patch Set 8:\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"22087ae9850f2109f106f47499c8517e34e0f1fd","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-06-18 00:00:47.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/8/20230618031231/NEC-V-FC : SUCCESS in 2h 51m 46s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/8/20230618031231/NEC-V-ISCSI : SUCCESS in 5h 48m 13s","accounts_in_message":[],"_revision_number":8},{"id":"c532679bf08846d65ef995a806a405b5f20a633e","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-06-20 12:44:05.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/8/20230620181235/NEC-FC : SUCCESS in 3h 30m 21s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/8/20230620181235/NEC-ISCSI : SUCCESS in 3h 31m 27s","accounts_in_message":[],"_revision_number":8},{"id":"c7e9829c0603d153ce73e34a7b568ea0faa65540","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-21 04:13:31.000000000","message":"Patch Set 8:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/90/862490/8/fc : SUCCESS in 0h 30m 22s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/90/862490/8/iscsi : SUCCESS in 0h 41m 21s","accounts_in_message":[],"_revision_number":8},{"id":"a602f14ee7224185d1887f243f01e67360a4b14d","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2023-07-04 08:33:53.000000000","message":"Patch Set 8:\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"06f62f126a5b869db1a0b6c0ea885ce68335c996","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-07-04 09:04:18.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9e12a788379704fb988f06fc9175d94b792784d2","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2023-07-04 09:17:21.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4295667720fa150e586e6781d8d143c2c812863c","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-07-04 09:22:49.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5682481fdbaf10a7497b5ff1bfe3b2c673a58642","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2023-07-04 09:26:46.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"02542739420e666bc24478866610fd1ada4922f2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-04 09:27:48.000000000","message":"Patch Set 9:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":9},{"id":"4139020fd0454810a846cc8eb0de57a9e711e612","author":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"date":"2023-07-04 09:28:20.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4a27fed8430f91e9510b4526351b0bab69f6dcd8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 09:36:55.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83bff48c2db2424f8abcde37fdb4308f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/523516912bfa48c39102f88a3185759d : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ac76fcc9af634fc88fce293051a14463 : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0f22262cde1f4f51af7c3cd467e88918 : SUCCESS in 9m 03s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"921d047b2c0eec711d3e2b0a0c6eb952cfde25ea","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-04 11:00:27.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-862490-9/ns5_nfs : FAILURE in 1m 31s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-862490-9/ns5_iscsi : FAILURE in 1m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"3dcf98b1d2b4c7dd7f7ddfc505530a308872e1dd","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-04 11:24:08.000000000","message":"Patch Set 9:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/30fd7f12d09447efb3c998139e8a013e\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/88c1e2cf28aa4358b46962b20ea0e716 : SUCCESS in 1h 14m 46s","accounts_in_message":[],"_revision_number":9},{"id":"7128bb7eb1941185b677a75219ff8d5201029ec2","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-07-04 13:00:09.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7377afdc11867c41e4b0bfa9df62d34f74e8943e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 13:16:54.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb8887b758e243c0a6b87aa5d338ca2f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68459968267f46038f30824c9edb5e22 : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5a600eb397e43a9a0cede1b7fcce38e : SUCCESS in 12m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c490e9483fa64b9cae9390925c5d70cd : SUCCESS in 12m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/148f2112a73d4524bb2c06e0ea533484 : SUCCESS in 12m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/580821a32ac74086a65229c2d2e4a928 : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/0b11ad0603c043478be77e53ee2378fe : SUCCESS in 48m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3692480638544ef1be97152d62a80021 : SUCCESS in 1h 59m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db7c62c8c5cf4cc98d1966f76f810e42 : SUCCESS in 53m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ec22c5b1863c48c0ab292a96d9828451 : SUCCESS in 19m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1a7c8671dc664c208944b0acd1942c1d : SUCCESS in 5m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/224156b0da3e49c5b1333d945da82d60 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cd5eb39ece294485a85e808bdd74dc2e : SUCCESS in 6m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/263d0e574a8144679be0e3b252358797 : SUCCESS in 6m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/76c14f3ace82492bbf144f5035bf0e08 : SUCCESS in 57m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4883a39743cc41f985598d4ca4cb39c9 : SUCCESS in 5m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c8cbd679dddc470aa69f2526421537c8 : SUCCESS in 1h 50m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/81d7903f11484928abcfd1b40251118e : FAILURE in 2h 00m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fafd28899c79428aac2e2eae11ad056b : SUCCESS in 1h 56m 14s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/36981f6c7771404ead6f0fdb8a600989 : SUCCESS in 2h 11m 36s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2c217b0754124a6093041d7a61d17e72 : SUCCESS in 1h 04m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b66d711258bc494f961c1c9e8d64ab8b : SUCCESS in 1h 24m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4446ea77923448eca1ca78c826553f29 : SUCCESS in 32m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c87bafb99d7940b3a362a4cebc0ae337 : SUCCESS in 1h 16m 30s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/cde7a19837ef4d6aa1aa915150d6fcc1 : SUCCESS in 1h 16m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/de219f7624804fd7b19c8d4b58cd606e : TIMED_OUT in 3h 03m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/906b5f158e74420bab286a00f300e16e : SUCCESS in 53m 50s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/3c6046dcaf044f458ef729b9d0b5e9f2 : SUCCESS in 12m 20s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1b798297c640f4065c436bbc382ff9212b805aed","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-04 14:22:57.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/90/862490/9/check/DellEMC_PowerFlex/30f0cec/DellEMC_PowerFlex/257 : FAILURE in 1h 50m 05s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"c2ab72c85fc5c1ef15b8177df8e3f8093723d497","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-04 15:00:08.000000000","message":"Patch Set 9:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/90/862490/9/fc : SUCCESS in 0h 49m 48s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/90/862490/9/iscsi : SUCCESS in 1h 40m 40s","accounts_in_message":[],"_revision_number":9},{"id":"b90e2a4758a5105715e6ce99a4dd3aa26ff61b61","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-04 16:47:03.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/90/862490/9/check/DellEMC_Unity_FC/0fc2514/DellEMC_Unity_FC/376 : SUCCESS in 1h 55m 53s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/90/862490/9/check/DellEMC_Unity_iSCSI/265c720/DellEMC_Unity_iSCSI/378 : SUCCESS in 1h 47m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"f840dbb9585d838bd1e101e58f57bb2135fabeb3","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-07-05 05:01:04.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3051e849afc1ebf6df190fa60a6d183e879337d9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-05 05:01:57.000000000","message":"Patch Set 9:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":9},{"id":"a3c375c19f307d9e2c064c7e7b18b46b53b31617","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 05:15:52.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ee523748c574118805588db0ddb2fe3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f807798b05294060920383c8633d853e : SUCCESS in 13m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/99300b11eff74e6088565dfea9f8ff54 : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2df893de968b4884aa2a6a5cdb52a9b1 : SUCCESS in 9m 28s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f9a01513f22c204955221e20ab70ffe585327b89","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-05 06:15:27.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-90-862490-9/FC/ : SUCCESS in 10h 16m 45s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-90-862490-9/iSCSI/ : SUCCESS in 9h 28m 32s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"17f62476efbb4591e456cf702bffb845fab3fd5a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-05 06:20:35.000000000","message":"Patch Set 9:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/8da4fb8c5ac844a5821f3d3d5296657c\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/2c68716f3a9343d5a02c1be503ca989d : SUCCESS in 1h 16m 32s","accounts_in_message":[],"_revision_number":9},{"id":"d11d6517994acd8407cdc54fd0b75ea12fe87d00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 08:42:44.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a4bf9cf1e6141cfbc40be5a7672d75d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac56f8c6edb44193aa87902781fe0e7f : SUCCESS in 8m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49fa48894ceb4525ab2d275dfc070cda : SUCCESS in 11m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cc42d2b67df4a00960f96e0c25a4750 : SUCCESS in 14m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4384c4a6dad4909a50c9dee709584f0 : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a9bbed6d0ff4792b3ba94906d0c477b : SUCCESS in 13m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/2b396b2aa54346519280abcdff8fd2d5 : POST_FAILURE in 1h 06m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de938fce836d45949fa7d18d59c3d204 : POST_FAILURE in 2h 08m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6e805146fda454c813b72cb117d48d9 : SUCCESS in 45m 13s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/59391c9f09f14453bbebfb17b18e5941 : SUCCESS in 16m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/70e0538ccaac40989f9e4bdb2f6ba110 : SUCCESS in 9m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c3bddbccd2f24e7cb5108341b6aaf1d7 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/76d3e0ae5a744cc38002f9df32557932 : SUCCESS in 5m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/658b522e1d0643349377b789b716fa8a : SUCCESS in 6m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/995d6fb3d2c94c8a8d8f1b584445e486 : SUCCESS in 33m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/50716cfa2d664c5a85cd7af7d686785e : SUCCESS in 5m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2959d5ff452c4423956c47cb574de68d : POST_FAILURE in 1h 03m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b3a1c0805a0d47de8940656be74c3c9e : POST_FAILURE in 1h 29m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1f385c61b9734f0f912116046373abbe : FAILURE in 2h 04m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/1d37816762254c46a14b21817e653e6d : POST_FAILURE in 2h 29m 10s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/63b9b749f5b24c4685343dd9abcbf404 : POST_FAILURE in 1h 12m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7e08e37788c641229d0e36776fa2a542 : SUCCESS in 1h 07m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/59dcb9dcb007466abb9e8207df6b7d71 : SUCCESS in 36m 35s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/990b420dc96c49abbaad511f74bfe9c3 : SUCCESS in 1h 37m 41s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2840a15ed99b4acf804bcd238d74918a : SUCCESS in 1h 46m 56s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eef7c83b5a8049c7b6e975b3e375cf64 : TIMED_OUT in 3h 04m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84dfc2c596744198adcdd2680c5b77a3 : SUCCESS in 1h 09m 09s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/a7d3c668e51441a8ae7ea9240090db88 : SUCCESS in 8m 51s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"3b23855652040857a73c2872c5258c8a9d27166e","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-06 12:06:28.000000000","message":"Patch Set 9:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/9/20230706151645/NEC-V-FC : SUCCESS in 3h 00m 44s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/9/20230706151645/NEC-V-ISCSI : FAILURE in 5h 49m 40s","accounts_in_message":[],"_revision_number":9},{"id":"e0c2bdf0384215f5dbcfc8918de1862847f42bbc","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-07 05:59:23.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/9/20230707064720/NEC-ISCSI : SUCCESS in 3h 45m 12s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/9/20230707064720/NEC-FC : SUCCESS in 8h 12m 00s","accounts_in_message":[],"_revision_number":9},{"id":"bcb4a7f15a82b7324b3dfc6168f53e7455486ac3","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2023-09-01 06:51:32.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c25a0bc3c9f8d8beb89bf5e8314a20eb4ecf4575","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-09-01 10:10:14.000000000","message":"Patch Set 9:\n\nBuild failed (FC). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/538/","accounts_in_message":[],"_revision_number":9},{"id":"9f915c1d03443a3f9abf9206bac356b2aa279a10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2023-09-01 10:20:24.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"479371c1017728da621fefe1b1ba0b3bf11d7bc5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 10:47:01.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9f26934901b45fb8f1d4c15f990d3ba\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a29585c81a4c4329b0b6052da445d23f : SUCCESS in 12m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7c2f39e4094948d3bc9e25c3ceb54498 : SUCCESS in 12m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/84dd51fccd0240dbb6dd2548a363165c : SUCCESS in 24m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7a4cff37f7064052a939a1e146825b3f : SUCCESS in 12m 02s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b3cc4abc5ed1a1827ebeb3bca00dafd58a74b444","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-01 10:48:07.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/90/862490/10/check/DellEMC_PowerFlex/eba7b73/DellEMC_PowerFlex/498 : FAILURE in 27m 24s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"6ddee755fb5ba542a0f2ffab0f6f9b064c121e32","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-01 11:25:44.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/224006d0f0ca4e0597362884ab2020ad : FAILURE in 29m 06s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/c5493fa4dc00443dbda6882e19f7c0d3 : FAILURE in 1h 02m 55s","accounts_in_message":[],"_revision_number":10},{"id":"f6521a534945e5b9beffa35ee190e810880ca333","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-01 11:32:03.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/90/862490/10/check/DellEMC_Unity_FC/9347659/DellEMC_Unity_FC/712 : FAILURE in 1h 11m 17s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/90/862490/10/check/DellEMC_Unity_iSCSI/6177f30/DellEMC_Unity_iSCSI/721 : FAILURE in 29m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"22fae870eb023f2a1694d78834ee1fc9cddac4b4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-09-01 11:41:21.000000000","message":"Patch Set 10:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/954ba099c309489fba488d052c25744d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/703b23260042431db84f6d0d130ca4c1 : FAILURE in 1h 18m 13s","accounts_in_message":[],"_revision_number":10},{"id":"f39612f8ac3b52a56d98d687f99840b4c73c85fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 12:17:21.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/972fa5a0bd594beb9c39ff715d8b4438\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeab36e2f3174d428f8a903dfac35767 : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecba7eb88d754152bc219f3a23d735d5 : SUCCESS in 11m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce966fc5beb34a8485fd0485914df77e : SUCCESS in 10m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/559f31ba90f2467baf4dc157645bd164 : SUCCESS in 10m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd407e550deb4e29983e5589fae2726b : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cd83523e37149d4a723e81c6b9a27ac : SUCCESS in 12m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/c57359b98ce043b5b3c46a62a5844f9a : SUCCESS in 41m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/17d1a2c04e8147b980492010fc2c672e : SUCCESS in 1h 26m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56f5903cb9ba44c789b4d9659bf46e71 : SUCCESS in 1h 03m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2bad2b21707d4bbb981b8af79f2279f7 : SUCCESS in 17m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/fe6919b40c534b80b6837786d1b612de : SUCCESS in 5m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e1e706d481ea4605875d7a14557129d5 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0999f879e4f44205bd07ce0413021717 : SUCCESS in 6m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/130790f8dfda45f9adeee12540a41d65 : SUCCESS in 5m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/29ec16dd68a84f339f560381423fec85 : SUCCESS in 30m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f9b08ae0c9ba420d8295421d986afd98 : SUCCESS in 5m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/19dd192cf1e44213a54d679c9ef26ed4 : SUCCESS in 1h 45m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/896d6d7dc62747418b91058e3a6a5403 : SUCCESS in 1h 49m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f1371c6ffb9a4c6aa8a7b44087cf7010 : SUCCESS in 1h 38m 51s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f659c2cbf5684ac7a40a387ffc1e71ca : SUCCESS in 1h 49m 28s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ad35ccaf5f584341a1dcfa40e4674ef4 : SUCCESS in 1h 12m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/025733caaaf049c193c3092c4f0d86a2 : SUCCESS in 1h 21m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a20d3b682b0749ee89f55b2f168cff83 : SUCCESS in 48m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/cd078fef805e427f9ada17d30e6b030c : SUCCESS in 1h 16m 14s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/2e0ff1b7d4684cf49f6d0abccdf42599 : SUCCESS in 1h 44m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/df536e351870405f8de19e25fb9a5707 : SUCCESS in 1h 34m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ff0350bbbf94718b8f45b274d5d7d1e : SUCCESS in 40m 47s","accounts_in_message":[],"_revision_number":10},{"id":"86094c71da10ce3d12ae7e999926031c762b1896","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-01 12:20:12.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/90/862490/10/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/0ae23bb/ : SUCCESS in 1h 58m 38s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/90/862490/10/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/fe052fe/ : SUCCESS in 1h 56m 22s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/90/862490/10/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/3677062/ : SUCCESS in 1h 37m 42s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/90/862490/10/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/0852290/ : SUCCESS in 1h 44m 38s","accounts_in_message":[],"_revision_number":10},{"id":"21b4f80cca04b0bf4057830ec2eb27f34584085f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-01 12:46:14.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/90/862490/10/check/DellEMC_PowerFlex/92c9bd3/DellEMC_PowerFlex/499 : FAILURE in 28m 35s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"051588032347555373ff2784f8bff422ecde2b2b","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-09-01 12:52:33.000000000","message":"Patch Set 10:\n\nBuild failed.","accounts_in_message":[],"_revision_number":10},{"id":"ccec1650397c49acc0d1ddbe50ee47cd8d4dd9fb","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-09-01 12:53:02.000000000","message":"Patch Set 10:\n\nBuild failed.","accounts_in_message":[],"_revision_number":10},{"id":"bc42222219d01252a885d79dc86823f6538cd51b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-01 12:58:28.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/90/862490/10/check/DellEMC_PowerStore_iSCSI/9a4ad39/DellEMC_PowerStore_iSCSI/252 : FAILURE in 31m 33s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/90/862490/10/check/DellEMC_PowerStore_FC/ab0cade/DellEMC_PowerStore_FC/261 : FAILURE in 40m 51s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/90/862490/10/check/DellEMC_PowerStore_NFS/c735a55/DellEMC_PowerStore_NFS/261 : FAILURE in 28m 56s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/90/862490/10/check/DellEMC_PowerStore_NVMeTCP/4be33d7/DellEMC_PowerStore_NVMeTCP/251 : FAILURE in 29m 51s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"95062e5894372f1a29f0e9218454f10cb4dbe561","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-01 13:17:18.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/862490/10/upstream-check/cinder-tempest-plugin-ontap-fcp/88a47b2/ : FAILURE in 44m 43s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/862490/10/upstream-check/cinder-tempest-plugin-ontap-iscsi/5cecd77/ : FAILURE in 53m 32s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/862490/10/upstream-check/cinder-tempest-plugin-ontap-nfs/0ccde58/ : FAILURE in 32m 04s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/862490/10/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/5b83412/ : FAILURE in 52m 45s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"3f639e7a1b9554a9f4b9d5998b3e7c90442e5064","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-01 13:24:13.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/90/862490/10/check/DellEMC_Unity_FC/06f78da/DellEMC_Unity_FC/713 : FAILURE in 1h 06m 32s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/90/862490/10/check/DellEMC_Unity_iSCSI/56114de/DellEMC_Unity_iSCSI/722 : FAILURE in 29m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"c086a79a762ade214a34772fa14dc63305372f17","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-09-01 14:00:07.000000000","message":"Patch Set 10:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/90/862490/10/fc : SUCCESS in 1h 02m 37s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/90/862490/10/iscsi : SUCCESS in 1h 11m 52s","accounts_in_message":[],"_revision_number":10},{"id":"6b9065055c1b0ea204564f11bfb061be87a46084","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-09-01 14:13:19.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-862490-10/ns5_nfs : SUCCESS in 3h 46m 54s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-862490-10/ns5_iscsi : SUCCESS in 3h 52m 8s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"5f05d65f06cd2d680db89ca09a63b9f63392b22e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-09-01 15:20:35.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1771/","accounts_in_message":[],"_revision_number":10},{"id":"1fa87358da5869c4891e96e1e8f3d3cbcc7dfbf1","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-09-02 04:04:54.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/10/20230902100456/NEC-V-ISCSI : SUCCESS in 2h 48m 29s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/10/20230902100456/NEC-V-FC : SUCCESS in 2h 59m 54s","accounts_in_message":[],"_revision_number":10},{"id":"eed3c9d8965ba4c837c55b2481d0b16220586415","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-09-02 10:08:05.000000000","message":"Patch Set 10:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-90-862490-10/FC/ : SUCCESS in 4h 45m 48s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-90-862490-10/iSCSI/ : FAILURE in 23h 47m 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":10},{"id":"723f08badee0cc1bcb297faa02a0785daa3f93b7","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-09-03 22:11:38.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/10/20230904030251/NEC-ISCSI : SUCCESS in 3h 47m 17s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/862490/10/20230904030251/NEC-FC : SUCCESS in 4h 08m 43s","accounts_in_message":[],"_revision_number":10},{"id":"aaa15ddacfb8a3cd447b478b75dae79f49bf5765","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2023-09-04 07:01:20.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"cf222201cb8e5baad69240aa92e9a3a7fa84fbe4","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-09-04 07:23:55.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"3af880681243c0536db15d5dc86ab1d76dfe433f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-09-04 15:17:35.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1778/","accounts_in_message":[],"_revision_number":10},{"id":"a1e25c3dfc64a2f6eaa21486eb65954326dfbdda","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-09-10 03:33:47.000000000","message":"Patch Set 10:\n\nBuild succeeded.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/90/862490/10/check/tyds_test/0b0c42b : SUCCESS in 3h 29m 55s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"600a87879a204c035747283a3c5fec4f6970e240","revisions":{"ea4d7bd572c7768da9a44c55215c40d8eca5ac17":{"kind":"REWORK","_number":1,"created":"2022-10-24 12:30:59.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/90/862490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/862490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/862490/1"}}},"commit":{"parents":[{"commit":"27bb0b01d65fe7d2c54a415eaeb012a0e9551680","subject":"Merge \"Remove forgotten variable _backup_db_fields\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27bb0b01d65fe7d2c54a415eaeb012a0e9551680"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-10-24 09:17:19.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-10-24 09:17:19.000000000","tz":180},"subject":"Tatlin driver - rename tatlin_api to tatlin_client","message":"Tatlin driver - rename tatlin_api to tatlin_client\n\nIn Tatlin Unified driver client object incorrectly is called\ntatlin_api. TatlinAPI is a different type and does not have\nTatlin*Client methods.\n\nImplements: blueprint yadro-tatlin-unified-refactoring\nChange-Id: I10135877889a17c6ae6274f8be1b992294bf65d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ea4d7bd572c7768da9a44c55215c40d8eca5ac17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ea4d7bd572c7768da9a44c55215c40d8eca5ac17"}]},"branch":"refs/heads/master"},"ec95a5624dc9cf744a37022e8b9ecc60423fe9be":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-10-24 12:58:51.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/90/862490/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/862490/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/862490/2"}}},"commit":{"parents":[{"commit":"27bb0b01d65fe7d2c54a415eaeb012a0e9551680","subject":"Merge \"Remove forgotten variable _backup_db_fields\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/27bb0b01d65fe7d2c54a415eaeb012a0e9551680"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-10-24 09:17:19.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-10-24 12:57:26.000000000","tz":180},"subject":"Tatlin unified driver - rename tatlin_api object","message":"Tatlin unified driver - rename tatlin_api object\n\nIn Tatlin Unified driver client object incorrectly is called\ntatlin_api. TatlinAPI is a different type and does not have\nTatlin*Client methods. Renamed to tatlin_client.\n\nImplements: blueprint yadro-tatlin-unified-refactoring\nChange-Id: I10135877889a17c6ae6274f8be1b992294bf65d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ec95a5624dc9cf744a37022e8b9ecc60423fe9be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ec95a5624dc9cf744a37022e8b9ecc60423fe9be"}]},"branch":"refs/heads/master"},"60f928d679f1c0a42595cfe2c7ffb20256c00cf8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-01-31 12:21:44.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/90/862490/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/862490/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/862490/3"}}},"commit":{"parents":[{"commit":"fce6d82a5d18281049817f92db194a38c61db5a8","subject":"Merge \"image_utils: Assume qemu-img \u003e\u003d 2.10.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fce6d82a5d18281049817f92db194a38c61db5a8"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-10-24 09:17:19.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2023-01-31 12:21:44.000000000","tz":0},"subject":"Tatlin unified driver - rename tatlin_api object","message":"Tatlin unified driver - rename tatlin_api object\n\nIn Tatlin Unified driver client object incorrectly is called\ntatlin_api. TatlinAPI is a different type and does not have\nTatlin*Client methods. Renamed to tatlin_client.\n\nImplements: blueprint yadro-tatlin-unified-refactoring\nChange-Id: I10135877889a17c6ae6274f8be1b992294bf65d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/60f928d679f1c0a42595cfe2c7ffb20256c00cf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/60f928d679f1c0a42595cfe2c7ffb20256c00cf8"}]},"branch":"refs/heads/master","description":"Rebase"},"976afd9844d3e57127738776939476d2cd377240":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-02-15 15:21:31.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/90/862490/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/862490/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/862490/4"}}},"commit":{"parents":[{"commit":"1a2b0796a2c2f3710af2d2b1d5a20004138b8f2b","subject":"Merge \"Doc: Branding change for Spectrum Virtualize family\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1a2b0796a2c2f3710af2d2b1d5a20004138b8f2b"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-10-24 09:17:19.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2023-02-15 15:21:31.000000000","tz":0},"subject":"Tatlin unified driver - rename tatlin_api object","message":"Tatlin unified driver - rename tatlin_api object\n\nIn Tatlin Unified driver client object incorrectly is called\ntatlin_api. TatlinAPI is a different type and does not have\nTatlin*Client methods. Renamed to tatlin_client.\n\nImplements: blueprint yadro-tatlin-unified-refactoring\nChange-Id: I10135877889a17c6ae6274f8be1b992294bf65d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/976afd9844d3e57127738776939476d2cd377240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/976afd9844d3e57127738776939476d2cd377240"}]},"branch":"refs/heads/master","description":"Rebase"},"de1ae1a4a6ddcdf1a2f2bec12f5dd68ea7368852":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-02-15 15:21:52.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/90/862490/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/862490/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/862490/5"}}},"commit":{"parents":[{"commit":"1a2b0796a2c2f3710af2d2b1d5a20004138b8f2b","subject":"Merge \"Doc: Branding change for Spectrum Virtualize family\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1a2b0796a2c2f3710af2d2b1d5a20004138b8f2b"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-10-24 09:17:19.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2023-02-15 15:21:52.000000000","tz":0},"subject":"Tatlin unified driver - rename tatlin_api object","message":"Tatlin unified driver - rename tatlin_api object\n\nIn Tatlin Unified driver client object is incorrectly called\ntatlin_api. TatlinAPI is a different type and does not have\nTatlin*Client methods. Renamed to tatlin_client.\n\nImplements: blueprint yadro-tatlin-unified-refactoring\nChange-Id: I10135877889a17c6ae6274f8be1b992294bf65d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/de1ae1a4a6ddcdf1a2f2bec12f5dd68ea7368852"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/de1ae1a4a6ddcdf1a2f2bec12f5dd68ea7368852"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9c0c7796f01db8b57c333d1e16e61c65326f6d3c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-03-01 16:31:32.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/90/862490/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/862490/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/862490/6"}}},"commit":{"parents":[{"commit":"9f7cb571bb95fec99f6113fc87785b3a271d5fc2","subject":"Merge \"Tests: Fix typo\u0027d unit test name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9f7cb571bb95fec99f6113fc87785b3a271d5fc2"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-10-24 09:17:19.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2023-03-01 16:31:32.000000000","tz":0},"subject":"Tatlin unified driver - rename tatlin_api object","message":"Tatlin unified driver - rename tatlin_api object\n\nIn Tatlin Unified driver client object is incorrectly called\ntatlin_api. TatlinAPI is a different type and does not have\nTatlin*Client methods. Renamed to tatlin_client.\n\nImplements: blueprint yadro-tatlin-unified-refactoring\nChange-Id: I10135877889a17c6ae6274f8be1b992294bf65d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9c0c7796f01db8b57c333d1e16e61c65326f6d3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9c0c7796f01db8b57c333d1e16e61c65326f6d3c"}]},"branch":"refs/heads/master","description":"Rebase"},"15050c6d8038cdbef27602296c52dfc715179002":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2023-03-10 12:08:10.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/90/862490/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/862490/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/862490/7"}}},"commit":{"parents":[{"commit":"9f7cb571bb95fec99f6113fc87785b3a271d5fc2","subject":"Merge \"Tests: Fix typo\u0027d unit test name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9f7cb571bb95fec99f6113fc87785b3a271d5fc2"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-10-24 09:17:19.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2023-03-10 12:08:10.000000000","tz":0},"subject":"Tatlin unified driver - rename tatlin_api object","message":"Tatlin unified driver - rename tatlin_api object\n\nIn Tatlin Unified driver client object is incorrectly called\ntatlin_api. TatlinAPI is a different type and does not have\nTatlin*Client methods. Renamed to tatlin_client.\n\nPartially Implements: blueprint yadro-tatlin-unified-refactoring\nChange-Id: I10135877889a17c6ae6274f8be1b992294bf65d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/15050c6d8038cdbef27602296c52dfc715179002"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/15050c6d8038cdbef27602296c52dfc715179002"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9a8b6a180a19ea12613930cfaa6dbd2c146c82bc":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-06-13 13:52:00.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/90/862490/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/862490/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/862490/8"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-10-24 09:17:19.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2023-06-13 13:52:00.000000000","tz":0},"subject":"Tatlin unified driver - rename tatlin_api object","message":"Tatlin unified driver - rename tatlin_api object\n\nIn Tatlin Unified driver client object is incorrectly called\ntatlin_api. TatlinAPI is a different type and does not have\nTatlin*Client methods. Renamed to tatlin_client.\n\nPartially Implements: blueprint yadro-tatlin-unified-refactoring\nChange-Id: I10135877889a17c6ae6274f8be1b992294bf65d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9a8b6a180a19ea12613930cfaa6dbd2c146c82bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9a8b6a180a19ea12613930cfaa6dbd2c146c82bc"}]},"branch":"refs/heads/master","description":"Rebase"},"fefcef0b390222cf3bc8ee06fc33020671cb1537":{"kind":"REWORK","_number":9,"created":"2023-07-04 09:26:46.000000000","uploader":{"_account_id":13671,"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","username":"vb"},"ref":"refs/changes/90/862490/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/862490/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/862490/9"}}},"commit":{"parents":[{"commit":"25455c476f2d918d550ddf6f209cdfc05a2cf420","subject":"Merge \"db: Remove the legacy \u0027migration_version\u0027 table\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/25455c476f2d918d550ddf6f209cdfc05a2cf420"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-10-24 09:17:19.000000000","tz":180},"committer":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2023-07-04 09:26:24.000000000","tz":180},"subject":"Tatlin unified driver - rename tatlin_api object","message":"Tatlin unified driver - rename tatlin_api object\n\nIn Tatlin Unified driver client object is incorrectly called\ntatlin_api. TatlinAPI is a different type and does not have\nTatlin*Client methods. Renamed to tatlin_client.\n\nPartially Implements: blueprint yadro-tatlin-unified-refactoring\n\nChange-Id: I10135877889a17c6ae6274f8be1b992294bf65d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fefcef0b390222cf3bc8ee06fc33020671cb1537"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fefcef0b390222cf3bc8ee06fc33020671cb1537"}]},"branch":"refs/heads/master"},"600a87879a204c035747283a3c5fec4f6970e240":{"kind":"REWORK","_number":10,"created":"2023-09-01 10:20:24.000000000","uploader":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"ref":"refs/changes/90/862490/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/862490/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/862490/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/862490/10"}}},"commit":{"parents":[{"commit":"4c9ffabe5c13ab00a27c240758afc8536d43472e","subject":"Merge \"Use openstack-tox-py311 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c9ffabe5c13ab00a27c240758afc8536d43472e"}]}],"author":{"name":"Vladislav Belogrudov","email":"v.belogrudov@yadro.com","date":"2022-10-24 09:17:19.000000000","tz":180},"committer":{"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","date":"2023-09-01 10:19:40.000000000","tz":180},"subject":"Tatlin unified driver - rename tatlin_api object","message":"Tatlin unified driver - rename tatlin_api object\n\nIn Tatlin Unified driver client object is incorrectly called\ntatlin_api. TatlinAPI is a different type and does not have\nTatlin*Client methods. Renamed to tatlin_client.\n\nPartially Implements: blueprint yadro-tatlin-unified-refactoring\n\nChange-Id: I10135877889a17c6ae6274f8be1b992294bf65d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/600a87879a204c035747283a3c5fec4f6970e240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/600a87879a204c035747283a3c5fec4f6970e240"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
