)]}'
{"id":"openstack%2Fcinder~863927","triplet_id":"openstack%2Fcinder~master~I73c37b005e0640a2482a4a314c62dba48438c53e","project":"openstack/cinder","branch":"master","topic":"nvme","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-07-30 19:35:30.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I73c37b005e0640a2482a4a314c62dba48438c53e","subject":"DNM: test LVM NVMe support config patch","status":"ABANDONED","created":"2022-11-07 19:25:59.000000000","updated":"2024-07-30 19:35:30.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"418475837c434c1f7b780ff146fc8514e74b7eb7","_number":863927,"virtual_id_number":863927,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-11-07 21:19:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-07 19:26:10.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-11-07 19:44:24.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-11-07 19:47:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-07 19:56:06.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2022-11-07 21:19:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-07 21:22:56.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-07 21:27:37.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-11-07 22:25:40.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-07 23:00:41.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-08 03:52:32.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-08 07:50:01.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-11-08 09:35:27.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-11-08 15:12:07.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2023-05-29 12:24:15.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"}],"messages":[{"id":"2e56976e21105abf4ef93124cb3b2a0b09d66c2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-11-07 19:25:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bde62d0cac0727c12e959cbcb6f157a4456d559","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-07 19:26:10.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1708 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"9d723dd220d5684499f9bf33c8de759530c761e1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-07 19:44:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : MERGER_FAILURE in 1m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1cdda04ac50ee6f30d30412d58a43b84ce889f92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 19:47:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7fe4ff04aac4e5887fe830ed5c21484\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4691655fd01a4777b6a4d1b05d4686b8 : SUCCESS in 18m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/67def8b3716f47a0894282d5fed2f728 : SUCCESS in 18m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"084c6cf6660c93f2904fe3f7d4125326bebb2785","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-11-07 19:56:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/ad8583d70c474405ad613e5e6c871e58 : POST_FAILURE in 29m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e1f3ba48b951e4ace38774a7ff0710b4e3d5fb09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 21:19:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c84a446fcc04e31a18540ee1c6b9397\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c610b28d52c4f2cb5975237099f6ad0 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/093ade632cba4930b9f84508ecdf8a2d : SUCCESS in 11m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a7e31da08ca4af382a8502ecd6a9753 : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acdbdbe7372f49a0ae8f631656ba85be : SUCCESS in 9m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/5ee8d83be8184c5cb7cc9d7e15787c86 : SUCCESS in 1h 13m 09s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/91d0b2d8c4d94349a03de232f426c64c : SUCCESS in 45m 15s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9238445b5ecc43629b8dd57a7ebe6e61 : SUCCESS in 1h 18m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0689fa9741db4de094b90c86961f7178 : SUCCESS in 40m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f35957947a4641d4afd3ec07ec29d954 : SUCCESS in 15m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c53089d19a9f459a9004eb2d52fa2769 : SUCCESS in 7m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a3a2fe5f9d214d72a025baeded445653 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b311ab82f94942cdbecfbec6d8ed8f37 : SUCCESS in 5m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/10a707ed3edd4e1eb0826e6d8a7520d6 : SUCCESS in 5m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7a4f000b353c45d7ade146745a3c40d5 : SUCCESS in 28m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d17e48837ec7462eace03aeca39f9f49 : FAILURE in 5m 20s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8bb1e9c147c8462a969d78023e4b881c : SUCCESS in 1h 10m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ee8f60563fa34da792b703137d43e9d2 : FAILURE in 1h 19m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d163511a390e4333a32da96209cb99c1 : FAILURE in 1h 30m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f631241976eb438190191aaa4507c402 : SUCCESS in 1h 02m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6d7fe7f09b3a4607a501b247bae090d1 : SUCCESS in 1h 13m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9b4072176da94effa99f00d7c1460dfb : SUCCESS in 36m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c9c919d73b8940ceadc8b7629960ad1a : SUCCESS in 1h 22m 33s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/932885171ca94fbd8b9adb932af444ea : SUCCESS in 1h 51m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ddcb493250c40b9b177092db38f4d97 : SUCCESS in 34m 24s","accounts_in_message":[],"_revision_number":1},{"id":"bf7d3e5d5ba3846bfb2505a713e89ffeb1a17c8a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-07 21:20:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/27/863927/1/check/iscsi-driver-alletra-6k/70f3847/70f3847.html : FAILURE in 1h 52m 34s","accounts_in_message":[],"_revision_number":1},{"id":"eb873d435c08eeaa91e135f549c7acc403220c7c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-07 21:22:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/863927 : SUCCESS in 1h 53m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a82261bfbff2a8ecc1ebdf2e892952cfd6f9a5e7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-07 21:27:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/27/863927/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/aa9f83c/ : SUCCESS in 1h 55m 29s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/27/863927/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/dbaf4d7/ : SUCCESS in 1h 30m 20s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/27/863927/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/b7c97fe/ : FAILURE in 2h 00m 27s","accounts_in_message":[],"_revision_number":1},{"id":"0d8c8a34ee7b96b707dcf848ebc21859f680fbb7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-07 21:36:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/863927 : FAILURE in 2h 07m 13s","accounts_in_message":[],"_revision_number":1},{"id":"aef0ef69e30328dd0dcfdeb39b99be63f6cc61c5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-11-07 22:25:40.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/7652240c05d545bea13b58ae8830dd3c : POST_FAILURE in 1h 52m 56s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/cee803bce76e4297badce7e48a90c624 : POST_FAILURE in 2h 24m 50s","accounts_in_message":[],"_revision_number":1},{"id":"eeea9173c5a40cebc049268e6dc5eebad001886c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-07 23:00:41.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-27-863927-1/ns5_nfs : SUCCESS in 3h 28m 17s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-27-863927-1/ns5_iscsi : SUCCESS in 3h 33m 56s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"4963b244d27ed1a7bfe347cf89ba5beaca6cc9d0","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-11-08 03:52:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/27/863927/1/20221108091500/NEC-FC : SUCCESS in 3h 35m 32s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/27/863927/1/20221108091500/NEC-ISCSI : SUCCESS in 3h 37m 28s","accounts_in_message":[],"_revision_number":1},{"id":"12dbd82b1e406217a402c86982b674fda3b08d51","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-08 07:50:01.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"2a224bd85d5871751c77b30677036790b634ea60","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-11-08 09:35:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/27/863927/1/20221108123143/NEC-V-ISCSI : SUCCESS in 2h 50m 21s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/27/863927/1/20221108123143/NEC-V-FC : SUCCESS in 6h 03m 40s","accounts_in_message":[],"_revision_number":1},{"id":"86415d201bcc76bef6aeacda3b19398456e49c77","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-11-08 15:12:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"434d5cccef8d85fabc72a64c5f984200c91a97bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 15:31:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6385da45fa64d53839686b74827fb11\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a477105e750c434c9bebeaeb3da8beb5 : SUCCESS in 14m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d9930197517249cc8aadd71ee2dfd009 : SUCCESS in 13m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"830260256d3cac881aa43bc30f1bc0785d585c7b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-11-08 15:42:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"94bb20e1bc95c94a8ce4684710a58995c7208d46","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-08 15:43:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 7m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c9d938fdeddf8892ab3b00112a9f70e17a058b12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 16:07:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0c966fc9a9143e6b94af62a31efeaa8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a7e56d30ff1f4db2ad0c0aa00f3f5f6e : SUCCESS in 21m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/515d266ab74b45488a04186144d46164 : FAILURE in 22m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d42356c3a322f2b4c4b42fca49f56d6d8af34d0","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-11-08 16:15:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/9f642a16b3a2474884ee54833faf9438 : FAILURE in 59m 22s","accounts_in_message":[],"_revision_number":1},{"id":"8d9b0ef1a9534fb41952b20c33362ef7dda621f2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-11-08 16:52:19.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/3cb5a5d98538470595658840a8824030 : POST_FAILURE in 1h 22m 40s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/8222abb9f4b84ce99e98a2cfeb4f3081 : POST_FAILURE in 1h 36m 24s","accounts_in_message":[],"_revision_number":1},{"id":"1ed382605db8669b30c442ec430e49c6277ad500","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 17:36:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c042e873bbc4a3fa161fb919df4cdca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8db27fbfa30842cdbe8d339c68884dd5 : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62a2b66a17444b6980b462e31bfe99c3 : SUCCESS in 8m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db420bca643f47d398ee89e686d01770 : SUCCESS in 10m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38504a582d0544589aab951c1874d8b4 : SUCCESS in 14m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/37123b61e91c48cb96c2d96f2441c202 : SUCCESS in 51m 02s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/077fcce7d51e425f855ece74037ee3a0 : SUCCESS in 59m 29s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/131763d2c94c42a79b6a832771915a46 : SUCCESS in 1h 10m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9998fe61b8634307a0c48f8141146477 : SUCCESS in 55m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7aacdd8c52574684b49f5b0e96545928 : SUCCESS in 15m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4f4374f94f604d8b89c344d3580a5c04 : SUCCESS in 5m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/538a38f7b96248faae3c0aae62826619 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ceee16e97ba24c50be8d92459501f796 : SUCCESS in 6m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2f018632b97c4f098db8e97b62942da8 : SUCCESS in 5m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d4534844aff84e028d05498601b330dd : SUCCESS in 40m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9cb59d1ec28046b6af12977af4a75b95 : FAILURE in 6m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/22ddbac7908d4dba871390d62c4e12df : SUCCESS in 1h 11m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1774ac345bba4181b86e3f76cf12a2ff : FAILURE in 1h 25m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f486c3225e46411f974a30c1f743465f : SUCCESS in 1h 30m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e4ccbbffe9814885808c5cba63065e34 : FAILURE in 47m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4415c587a52348298a3b5528f8fa848f : SUCCESS in 1h 31m 55s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7e4575c948a3412499a1682bf15214c1 : SUCCESS in 42m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/53141b5ed1a544ffb8e63d6a8143f265 : SUCCESS in 1h 23m 16s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/de8fe25e453c45909d23c44da7458e41 : SUCCESS in 2h 16m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc00a1d37e6448229639ef0f2e021862 : SUCCESS in 45m 57s","accounts_in_message":[],"_revision_number":1},{"id":"d6e728df68f8a2e8203ecefd1cc300189fd96b77","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2023-05-29 12:24:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"418475837c434c1f7b780ff146fc8514e74b7eb7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-07-30 19:35:30.000000000","message":"Abandoned\n\nI don\u0027t even remember what this was for!","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d594fb880a7b437636d37fcb1287a295c6f368f3","revisions":{"d594fb880a7b437636d37fcb1287a295c6f368f3":{"kind":"REWORK","_number":1,"created":"2022-11-07 19:25:59.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/27/863927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/27/863927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/863927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/863927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/27/863927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/27/863927/1"}}},"commit":{"parents":[{"commit":"2603af55e18cdfbca930c1502e83fc05c35d5daf","subject":"Merge \"api-ref: Add docs for clusters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2603af55e18cdfbca930c1502e83fc05c35d5daf"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-11-07 19:23:43.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-11-07 19:23:43.000000000","tz":-300},"subject":"DNM: test LVM NVMe support config patch","message":"DNM: test LVM NVMe support config patch\n\nMake sure the devstack patch supports the current config\noptions.\n\nDepends-on: https://review.opendev.org/c/openstack/devstack/+/814193\nChange-Id: I73c37b005e0640a2482a4a314c62dba48438c53e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d594fb880a7b437636d37fcb1287a295c6f368f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d594fb880a7b437636d37fcb1287a295c6f368f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
