)]}'
{"id":"openstack%2Fcinder~864100","triplet_id":"openstack%2Fcinder~master~Iafa92ece7f83323af257c1702df5029469b11739","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-11-16 03:05:40.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2022-11-16 13:38:17.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2022-11-16 13:38:17.000000000","reason":"Change was submitted"},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2022-11-16 12:12:59.000000000","reason":"\u003cGERRIT_ACCOUNT_20813\u003e replied on the change","reason_account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}}},"hashtags":[],"change_id":"Iafa92ece7f83323af257c1702df5029469b11739","subject":"Set backup status to error on VolumeNotFound","status":"MERGED","created":"2022-11-09 10:15:56.000000000","updated":"2022-11-16 16:38:52.000000000","submitted":"2022-11-16 13:38:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864100","meta_rev_id":"377e6f8460834945a961b46586ebaa0189b66376","_number":864100,"virtual_id_number":864100,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-16 13:38:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-11-16 03:05:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-16 12:12:59.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-16 12:12:59.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-09 10:16:03.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-11-09 10:21:13.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-11-09 10:24:42.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-11-09 10:35:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-09 11:20:34.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2022-11-09 11:52:08.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2022-11-09 11:56:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-09 11:58:01.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-09 12:10:16.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-11-09 12:14:41.000000000","updated_by":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"reviewer":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"state":"CC"},{"updated":"2022-11-09 12:18:07.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-09 12:41:24.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-11-09 12:50:32.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-11-09 13:01:10.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-09 13:10:09.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-11-09 13:29:27.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-09 13:48:26.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-09 15:17:26.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2022-11-09 15:26:00.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-11-09 19:55:16.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-11-09 21:30:21.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-10 13:41:44.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-11-16 14:12:18.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"}],"messages":[{"id":"3f412b94407cc7a5a8ae3e639b9bd49c590cb280","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-11-09 10:15:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2d7664ec5a7cb466f82fe550344d73dc13ef4d8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-09 10:16:03.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1713 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"7fe6a094901475fd240fd12990d7bf737e5a1425","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-09 10:21:13.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"3c11f10aca1de7316ead53d070e59345995353f7","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-09 10:24:42.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"ce5007c0a2b9c6142b5ec277487121092e36eb2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 10:35:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/047713268bb1419684330f5fa5302ecd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/41094f099f094e668a2bf1bfd0b65b35 : SUCCESS in 15m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d5e80f7b87364d439d54a6b5fb70f19d : SUCCESS in 14m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e144e22599ec111f46264bbfbd30bb643551720a","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2022-11-09 11:20:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/841621a75a724aceb787d01cea2204ec : FAILURE in 1h 04m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a790846be49a7f7995c9757354f439a1bfa8db42","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-11-09 11:52:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"db75586f7188a0475ce824b1aa74aa1c1fffabc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 11:56:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d38f21149f344bae8fde774c4471e7fb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/039e7baa89d74431ac603c64ce8fee9d : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acf2d352e9bd4bd18ecae82f03bdfa1d : SUCCESS in 10m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4313bd47641b4c7ebc3725e5f98a09a4 : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c21fc6e5ade414b824453494076101e : SUCCESS in 12m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/354e600409674b3f9da4475324357672 : SUCCESS in 45m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bbc058f24ba149da8347e8492af572d1 : SUCCESS in 49m 58s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a995f39eaa394f60b27290aae7c993a9 : SUCCESS in 58m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f57eeca1aebe475c91be22231002f7ee : SUCCESS in 46m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2920e7131cc434f8a78225bb7197119 : SUCCESS in 7m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a4f5de6679a14aa4b6050b68bd10f28f : SUCCESS in 13m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e42245a6c0204807b0dc0dc783b2017c : SUCCESS in 5m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/16c2d6fa4a4b4f5490b306fd01414730 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4abd2ce1806b4ca5b3578673d2291325 : SUCCESS in 6m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/271691e7773743019feed76b57691826 : SUCCESS in 5m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/15584c3b30724f41913aa8ecb475a8c5 : SUCCESS in 32m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/668cb4f59fc641bab7a355771bb6d474 : FAILURE in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5200cf8acc3f452c9fe4cc46ae68ac42 : SUCCESS in 1h 09m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dc5f7baabac14ec3a6d57f862d9f99a7 : FAILURE in 1h 22m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2e27fbb33afd4ef8b90c3c250a5b15fb : SUCCESS in 1h 21m 25s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/10d8d5da4c904bd1af803fb1e114900d : SUCCESS in 1h 02m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1886b78ef11140cda1532b471b39aa04 : SUCCESS in 1h 19m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/06303438a2bc48669c8a8bf2b1478532 : SUCCESS in 31m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/463844e2bac141649ecfb9ae2c75f5c3 : SUCCESS in 1h 15m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/72f5d2d1fefc4dbe99dc1fba48e74de4 : SUCCESS in 1h 38m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80c7032ab4784bc58df8c2ad6b97cd67 : SUCCESS in 39m 13s","accounts_in_message":[],"_revision_number":1},{"id":"f4856e5f236b15ea14bdc3d31d5a00522be1e7ce","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-11-09 11:58:01.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/ce35d42ddd5048c8821b2eb023e9be7b : POST_FAILURE in 1h 33m 00s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/c070a587561b480786eee5a59fbccb12 : POST_FAILURE in 1h 39m 14s","accounts_in_message":[],"_revision_number":1},{"id":"3147d3366114c2365ca0d9b7b38f43149fd8b5f1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-09 12:10:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/00/864100/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/22339ce/ : SUCCESS in 1h 51m 53s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/00/864100/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/bd5c408/ : SUCCESS in 1h 35m 49s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/00/864100/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/7f27cf3/ : SUCCESS in 1h 32m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a40cd319f7e07e16fc794baf2f4b7c60da176eab","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-11-09 12:14:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-00-864100-1/1667994991 : SUCCESS in 18m 07s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":1},{"id":"51f6bc590fab5fc0eede7ad315d75b75b293fc1e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-09 12:18:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/864100 : SUCCESS in 1h 58m 11s","accounts_in_message":[],"_revision_number":1},{"id":"9f31b7f999451b8fb8fe0c8efbaaa7023e688cb5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-09 12:39:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/864100 : FAILURE in 2h 17m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b4bf18511a18b6b070e6fe590d4add94c4597cbe","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-11-09 12:41:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/864100/1/upstream-check/cinder-tempest-plugin-ontap-fcp/0928e1a/ : RETRY_LIMIT in 7m 52s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/864100/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/f00d49d/ : SUCCESS in 43m 45s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/864100/1/upstream-check/cinder-tempest-plugin-ontap-nfs/63a6226/ : SUCCESS in 42m 17s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"d6213c7a5a0ccf131e42b6a6f6d20603c61c3b08","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-11-09 12:50:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/864100/1/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/3cafd18/ : SUCCESS in 52m 32s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"7e6b3786f6dbccf5d47591986d68a15282a56644","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-11-09 13:01:10.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/00/864100/1/check/DellEMC_PowerFlex/1b9cade/DellEMC_PowerFlex/1456 : FAILURE in 1h 04m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"913c08ec9b37c3f2095c4ab77810535716edf5a1","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-11-09 13:10:09.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/00/864100/1/check/DellEMC_SC_FC/f2b7976/DellEMC_SC_FC/1097 : SUCCESS in 48m 21s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/00/864100/1/check/DellEMC_SC_iSCSI/75bdc00/DellEMC_SC_iSCSI/1081 : SUCCESS in 1h 13m 20s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"3fee6d0fec9ec2ac9fb81d7c6feb1b477ace8915","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-09 13:21:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/00/864100/1/check/iscsi-driver-alletra-6k/d9ceb1e/d9ceb1e.html : TIMED_OUT in 3h 01m 28s","accounts_in_message":[],"_revision_number":1},{"id":"05dd61b63c0cbc373142519ac9a895bdb8b3de52","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-11-09 13:29:27.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/00/864100/1/check/fujitsu-eternus-dx-iscsi/b1d3b87 : SUCCESS in 1h 25m 07s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/00/864100/1/check/fujitsu-eternus-dx-fc/f6f8b87 : SUCCESS in 1h 27m 17s","accounts_in_message":[],"_revision_number":1},{"id":"4e1275862f17b5e36eff502bb2d0b8ac2a196aa8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-09 13:48:26.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-00-864100-1/ns5_nfs : SUCCESS in 3h 29m 32s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-00-864100-1/ns5_iscsi : SUCCESS in 3h 31m 58s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"70e4bda56509bb0c8b43fadf7c8fbf3c46904710","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-11-09 15:26:00.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"33fe7fbfd955eaf2ae4eb806100844589d803871","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-11-09 19:55:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/864100/1/20221109230004/NEC-V-ISCSI : SUCCESS in 2h 43m 20s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/864100/1/20221109230004/NEC-V-FC : SUCCESS in 5h 55m 09s","accounts_in_message":[],"_revision_number":1},{"id":"67b38514d4a2a96f2fbbfeec07d5994cc2439b7d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-11-09 21:30:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/864100/1/20221109205223/NEC-FC : SUCCESS in 3h 08m 25s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/864100/1/20221109205223/NEC-ISCSI : SUCCESS in 9h 37m 55s","accounts_in_message":[],"_revision_number":1},{"id":"80c9ac2b3fd8f92966a78324400117f7ced3a193","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-11-10 11:43:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e915958af3187f96e32e3a95da5dbe51d851e6bc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-10 11:43:41.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1718 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"156875f17f402b24077997cdec2914892079f823","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-11-10 11:44:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8e3715a7b748e23192aa6595161489e57b510b66","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-11-10 11:44:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c2fdaf95686da4c136c4bcfc0ffd08dd7623f59c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-10 12:05:12.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 10m 28s","accounts_in_message":[],"_revision_number":2},{"id":"8bf4a041d787fbbe40e5711a9f2598539303cf44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 12:05:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af8992f0d36f49b9ace164758ff84321\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c0be6d4d4a241318d430ba93082a7f9 : SUCCESS in 17m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4a1563a66d62485a8622c0bb2a4e0dda : SUCCESS in 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a22996cdff8152c26b8f9863ab0882b73dfe825b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-10 12:51:11.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/00/864100/2/check/iscsi-driver-alletra-6k/c21ce93/c21ce93.html : FAILURE in 1h 06m 43s","accounts_in_message":[],"_revision_number":2},{"id":"3fd8410efcaafdabe54073f35ac585086a66e92a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-11-10 13:21:31.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/3d1f7007e32d40c6a361968f3388c6e6 : POST_FAILURE in 1h 16m 25s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/52d94c54957d49a0a6e82dbce5925d2e : POST_FAILURE in 1h 35m 13s","accounts_in_message":[],"_revision_number":2},{"id":"1af4d4cbb12b3d8ceb3ba1c119c724b8e387b805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 13:33:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60820b1eebb7452e9c70aa5ea7fc737f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e45086adff9f46c9bff237873e6dac9a : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/552b12dea1874143936e02d51ecc195a : SUCCESS in 8m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e5b667ffa4140aabe7e93f4d53b062d : SUCCESS in 9m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aab0486598540fc8a710164a38bed39 : SUCCESS in 10m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/495749ea44f84b27bb5d0931b572d57a : SUCCESS in 37m 58s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/73745eca9dac48b2ba1306ee817ffd4e : SUCCESS in 45m 11s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/878a2cd467a54f029d64d8d01a76204a : SUCCESS in 1h 19m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b503100671c4fb38d5420ce340960b5 : SUCCESS in 41m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3c06ce62a8744cb84563b11c3c56073 : SUCCESS in 7m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/29cceb2f4d654d779c5c0efed0a92da7 : SUCCESS in 11m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f7a1542606e8403cb8cf24c543345b2d : SUCCESS in 5m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a8d223395336408fba5b01ded506095c : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b904b23c30e84e1caaa9c7110c5953f7 : SUCCESS in 6m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fdddeaef8a224a2782d383522adcc52b : SUCCESS in 6m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ed4b930e0f4c48babc4883f1092680a9 : SUCCESS in 32m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b0e9944785dd43309d2426633ad5bd90 : FAILURE in 6m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8f82209b8db647139a4c0f89c0f8bc95 : SUCCESS in 1h 05m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f9c128679045479393a6de5ddc13b82a : FAILURE in 1h 16m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d5d810bb7aa74285abe3ae90328b3451 : SUCCESS in 1h 34m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b729647cb82348459a9e5edfce4aebf4 : SUCCESS in 1h 00m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b7e1a6b3816947f59a7d44aff91e4822 : SUCCESS in 1h 14m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/30252b7d57934c0baeb9eda50227bc33 : SUCCESS in 42m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/72dedf8832794bdfa65afec111a961a8 : SUCCESS in 1h 05m 24s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b8ddd50f277d4fe386949547a946af4d : SUCCESS in 1h 46m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b4b8ff6fe6c429cbd6cbda92068495d : SUCCESS in 56m 18s","accounts_in_message":[],"_revision_number":2},{"id":"ac8354cf226a6d6eca1ff0606097782b95b889da","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-11-10 13:41:44.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/10_11_2022_13_33_58\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"d3bb3ac78fd3d98be351b5dd5df093f474ba106b","author":{"_account_id":33000,"name":"Charles Piercey","display_name":"KIOXIA CI","email":"chuck.piercey@kioxia.com","username":"charlespiercey"},"date":"2022-11-10 13:53:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n* kumoscale-nvmeof-tempest http://104.254.65.37/refs-changes-00-864100-2/1668087227 : SUCCESS in 19m 39s\n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/KIOXIA_CI","accounts_in_message":[],"_revision_number":2},{"id":"96d80c9c29945f187d475563db101beb8ef977d2","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-11-10 13:54:30.000000000","message":"Patch Set 2:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/00/864100/2/check/fujitsu-eternus-dx-iscsi/5c30ba4 : SUCCESS in 1h 58m 46s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/00/864100/2/check/fujitsu-eternus-dx-fc/3f360d5 : SUCCESS in 2h 10m 19s","accounts_in_message":[],"_revision_number":2},{"id":"334973408abb6a95463509b3a8955fd7813d3210","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-10 14:07:13.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/864100 : SUCCESS in 2h 20m 21s","accounts_in_message":[],"_revision_number":2},{"id":"c4fc23a1a9eb6363a1a8f81e5602e82342e5066b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-10 14:14:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/864100 : SUCCESS in 2h 26m 52s","accounts_in_message":[],"_revision_number":2},{"id":"ce7024a4a797f8c5c6448978b4106738df82d511","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-11-10 14:18:23.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/864100/2/upstream-check/cinder-tempest-plugin-ontap-fcp/bcfae84/ : RETRY_LIMIT in 7m 56s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/864100/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/6070e48/ : FAILURE in 42m 43s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/864100/2/upstream-check/cinder-tempest-plugin-ontap-nfs/df77c29/ : SUCCESS in 43m 44s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"473c53793b7c318fb7b36ef7a89418d570c4e31a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-10 14:21:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/00/864100/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/7b01d39/ : SUCCESS in 1h 29m 34s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/00/864100/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7381d0e/ : SUCCESS in 1h 50m 12s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/00/864100/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/5f89933/ : FAILURE in 2h 36m 41s","accounts_in_message":[],"_revision_number":2},{"id":"87759a240ab487a07855f38edab735b68cd3053b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-11-10 14:34:09.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/00/864100/2/check/DellEMC_PowerFlex/f0b8e47/DellEMC_PowerFlex/1460 : SUCCESS in 1h 00m 05s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"1fdd1484b3e85e8615d9a7016ff12d032c97ce12","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-11-10 14:47:24.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/00/864100/2/check/DellEMC_SC_FC/f2967d3/DellEMC_SC_FC/1101 : SUCCESS in 48m 59s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/00/864100/2/check/DellEMC_SC_iSCSI/cfef0b1/DellEMC_SC_iSCSI/1085 : SUCCESS in 1h 13m 20s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"5f15fc7863f288cc5de89893d252a8a21a037892","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-10 14:49:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/00/864100/2/check/iscsi-driver-nimble/76eb37b/76eb37b.html : TIMED_OUT in 3h 00m 45s","accounts_in_message":[],"_revision_number":2},{"id":"190bfc090910f6c05347bf24533c1026fa208d17","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-11-10 14:55:10.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/864100/2/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/a009154/ : FAILURE in 1h 19m 36s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"06601038eb43a44a26c2a42ee24556f2f40781dd","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-10 15:13:44.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-00-864100-2/ns5_nfs : SUCCESS in 3h 25m 4s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-00-864100-2/ns5_iscsi : SUCCESS in 3h 29m 44s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"7f567def5348ae9262f9fd2f0f185876969330a5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-11-10 17:29:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/864100/2/20221110204334/NEC-V-ISCSI : SUCCESS in 2h 30m 36s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/864100/2/20221110204334/NEC-V-FC : SUCCESS in 5h 46m 12s","accounts_in_message":[],"_revision_number":2},{"id":"1e5b844d0f9c1b09829e2c0a280f893ef2b33a10","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-11-10 18:31:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/00/864100/2/20221110204332/NEC-FC : SUCCESS in 3h 32m 53s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/00/864100/2/20221110204332/NEC-ISCSI : SUCCESS in 6h 47m 56s","accounts_in_message":[],"_revision_number":2},{"id":"5996555d6ed338932c9c3b7c20ee2bc2b433fb0c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-11-16 03:05:40.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"167ed9481737e6c94ae30bc3809043dfad70f2f7","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-11-16 12:12:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1180e1653e464128245c75c409597e661dbd5e0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 12:13:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"566f8e0e4bfac622a66ee682f600599545c5eb8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 13:38:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b532fa9e9cf4ad7b059fbd65b50eda7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff23552c22ac4db7911b4fa32cc9f1cc : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/179af92382fe4ecbaf187c396144620f : SUCCESS in 9m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/981a028816ae4d35a8035bd29d3c2afd : SUCCESS in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2d03297c6ed482592eb6f180221ca06 : SUCCESS in 10m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/c4b0a043ef7847f59532b8ec3598332d : SUCCESS in 40m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/63d6437e63af42c99fad0dc086975beb : SUCCESS in 58m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00366dcc7b87469b90c256ff1f6204c7 : SUCCESS in 57m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e55435fea15404480bce78bf3ac5a50 : SUCCESS in 7m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a9e880b183174d68b4f3f3f5b9c98954 : SUCCESS in 46m 28s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/780070557d044bffabade7908c8d9658 : SUCCESS in 1h 18m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/338fb57268ed4e6ab37cec0880df76f8 : SUCCESS in 56m 09s","accounts_in_message":[],"_revision_number":2},{"id":"cb80e3bda43e8169360bf14dc65fc063e38658ef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 13:38:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4fbddacdf8bd3c5df9d432300452487335bf8304","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 13:39:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/207904c150d24d95b0c107598c46cea7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd6de9019f2440cd9b2e0052ddac53e6 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a7afb941d22471e9689afcfe6efc03e : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2},{"id":"422d64a55f734159ba2c354e7219ab73dee8c2b1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-11-16 14:12:18.000000000","message":"Patch Set 2:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/833 \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2},{"id":"377e6f8460834945a961b46586ebaa0189b66376","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-11-16 16:38:52.000000000","message":"Patch Set 2:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"813df9b99f23bdc54787a0c918e47f9d61df3232","revisions":{"c157021b61781532131a932d8d4741917f41cc38":{"kind":"REWORK","_number":1,"created":"2022-11-09 10:15:56.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/00/864100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/864100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/864100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/864100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/864100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/864100/1"}}},"commit":{"parents":[{"commit":"2603af55e18cdfbca930c1502e83fc05c35d5daf","subject":"Merge \"api-ref: Add docs for clusters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2603af55e18cdfbca930c1502e83fc05c35d5daf"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-11-09 10:14:03.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-11-09 10:14:03.000000000","tz":60},"subject":"Set backup status to error on VolumeNotFound","message":"Set backup status to error on VolumeNotFound\n\nWhen creating a backup and it fails we will try\nto update the status on the volume but if it does\nnot exist we will instead fail on that and the\nbackup will be stuck in creating status which means\ncloud admin need to help resetting the status.\n\nThis catches the VolumeNotFound exception and ignores\nupdating the volume status and thus backup will be\nset to error and can be deleted by user instead.\n\nCloses-Bug: #1996049\nChange-Id: Iafa92ece7f83323af257c1702df5029469b11739\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c157021b61781532131a932d8d4741917f41cc38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c157021b61781532131a932d8d4741917f41cc38"}]},"branch":"refs/heads/master"},"813df9b99f23bdc54787a0c918e47f9d61df3232":{"kind":"REWORK","_number":2,"created":"2022-11-10 11:43:31.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/00/864100/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/00/864100/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/864100/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/864100/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/00/864100/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/00/864100/2"}}},"commit":{"parents":[{"commit":"dcec2f6f01ffc63dc058f641370e9f5a0bad07e4","subject":"Merge \"Add doc8 check to docs builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dcec2f6f01ffc63dc058f641370e9f5a0bad07e4"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-11-09 10:14:03.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-11-10 11:43:23.000000000","tz":0},"subject":"Set backup status to error on VolumeNotFound","message":"Set backup status to error on VolumeNotFound\n\nWhen creating a backup and it fails we will try\nto update the status on the volume but if it does\nnot exist we will instead fail on that and the\nbackup will be stuck in creating status which means\ncloud admin need to help resetting the status.\n\nThis catches the VolumeNotFound exception and ignores\nupdating the volume status and thus backup will be\nset to error and can be deleted by user instead.\n\nCloses-Bug: #1996049\nChange-Id: Iafa92ece7f83323af257c1702df5029469b11739\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/813df9b99f23bdc54787a0c918e47f9d61df3232"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/813df9b99f23bdc54787a0c918e47f9d61df3232"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
