)]}'
{"id":"openstack%2Fcinder~864185","triplet_id":"openstack%2Fcinder~stable%2Fwallaby~Ib04e49dac9fdabd81809ca389553ce3638bb9bbf","project":"openstack/cinder","branch":"stable/wallaby","topic":"pure_iscsi_cidrs-stable/wallaby","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-11-22 08:45:46.000000000","reason":"Change was submitted"},"34598":{"account":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"last_update":"2022-11-22 08:45:46.000000000","reason":"Change was submitted"},"5689":{"account":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"last_update":"2022-11-18 03:02:40.000000000","reason":"\u003cGERRIT_ACCOUNT_5689\u003e replied on the change","reason_account":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"}}},"hashtags":[],"change_id":"Ib04e49dac9fdabd81809ca389553ce3638bb9bbf","subject":"[PURE] support IPv6 / add parameter pure_iscsi_cidr_list","status":"MERGED","created":"2022-11-14 14:35:18.000000000","updated":"2022-11-22 08:47:09.000000000","submitted":"2022-11-22 08:45:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864185-pure_iscsi_cidrs-stable/wallaby","cherry_pick_of_change":774283,"cherry_pick_of_patch_set":9,"meta_rev_id":"beada116558f525f686f38d6ee139387123c0b9a","_number":864185,"virtual_id_number":864185,"owner":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-22 08:45:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-11-21 15:39:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-11-18 03:02:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-18 18:34:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2022-11-21 19:48:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":1,"date":"2022-11-21 19:48:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33082,"name":"Christophe Drevet","display_name":"dr4Ke","email":"cdr@antemeta.fr","username":"dr4Ke"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-14 14:35:18.000000000","updated_by":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"reviewer":{"_account_id":33082,"name":"Christophe Drevet","display_name":"dr4Ke","email":"cdr@antemeta.fr","username":"dr4Ke"},"state":"CC"},{"updated":"2022-11-14 14:35:31.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-11-14 14:40:41.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-11-14 14:50:46.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-11-14 14:54:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-14 15:00:07.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-14 15:34:44.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-14 16:23:44.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-14 16:31:39.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-11-14 16:33:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-15 01:47:06.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"CC"},{"updated":"2022-11-15 02:37:38.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-17 16:12:50.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2022-11-17 22:17:25.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2022-11-17 22:26:57.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"CC"},{"updated":"2022-11-18 03:02:40.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2022-11-18 03:13:48.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-11-18 18:34:56.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2022-11-21 15:39:09.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-11-21 19:48:40.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"19dfc65f46aca6d26b689a1f43cc34c99f297824","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"date":"2022-11-14 14:35:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ebb51bbf117b0655f9501478892df59a1c679e3a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-14 14:35:31.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1725 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"2481952fed7fc5c7f2fc376d39b6683c67351070","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-14 14:40:41.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"921bc08c09de6fdb9509f1520f7fac6d91043335","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-14 14:50:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 13m 03s","accounts_in_message":[],"_revision_number":1},{"id":"35487a2948ddf2d0d37870738d85bc6ffdab3209","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 14:54:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/710523e8aee245018b21f61905e5c9c5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e66c8a032594bd3968b424f154ccb33 : SUCCESS in 10m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/510ef850bda645aea15994fa073a7077 : SUCCESS in 14m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"efd0856a16878d46d2e1e3f92ccb5b94fb288979","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-14 15:00:07.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-85-864185-1/ns5_nfs : FAILURE in 24m 16s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-85-864185-1/ns5_iscsi : FAILURE in 24m 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"3a766bc3ea07165aad2a26a1532aec0c4927693c","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-11-14 15:34:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/85/864185/1/ : FAILURE in 51m 13s","accounts_in_message":[],"_revision_number":1},{"id":"2229955a9380018dbf2260f5f7240d891c66b7ce","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-14 16:23:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/864185 : FAILURE in 1h 24m 44s","accounts_in_message":[],"_revision_number":1},{"id":"b1d69d85a7fdc0fe9e0d623c092cf425e16cb143","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-14 16:31:39.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/85/864185/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2de6c12/ : SUCCESS in 1h 55m 13s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/85/864185/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7a2ee86/ : SUCCESS in 1h 44m 25s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/85/864185/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/58ce9dc/ : FAILURE in 32m 03s","accounts_in_message":[],"_revision_number":1},{"id":"05f8d35871aa808033052960022a85c662a8da8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 16:33:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/77708d8813ab4cbbbd848d92c2fd1b2e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b768c56f4b4646b196048561379d171b : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1462061f368043d5956bc77e6b6830be : SUCCESS in 8m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d4a1ed907684d9eb3083e1ddc6cbae6 : SUCCESS in 8m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2aeae746b4ec43f887be404e3ca35e69 : SUCCESS in 9m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceeef730975247a28d81c102d3d657c1 : SUCCESS in 10m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/002f1658821c49c4a3771f43807e862d : SUCCESS in 57m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/833ab6066a7e4079870a9206f388d645 : SUCCESS in 1h 12m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d67d86db4a6f483eb88a28abeaf8cf73 : SUCCESS in 48m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2ebbbb764e748fca11877b4e5d952bf : SUCCESS in 7m 42s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f05b4c9a64fd4308be1e7eb9ee42e985 : SUCCESS in 11m 45s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/08a8fe2dd89248718fb60dde45ee5ed8 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ea67ce10fc734b3bb73155f1c3790298 : SUCCESS in 5m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8b5bb173e41b4bd09400a0d84035f369 : SUCCESS in 5m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4fdf68560f65479fbc7dfaf14e2c48bc : SUCCESS in 36m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/883ec1ca039a42518db8dd21c55c8fb4 : FAILURE in 5m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/98ab0c14aa6748a28fafdea1782a4232 : FAILURE in 1h 11m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/ccf096231b344a149906b7fdaa1800d8 : FAILURE in 1h 26m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b94c91192ea748a4926fd4a719b5f3c7 : SUCCESS in 1h 29m 53s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dcc97a63c8fe486ba44b89659b31211f : SUCCESS in 1h 07m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7b135a134971410ba8fc57e2b5c7793a : SUCCESS in 1h 26m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fbeba480d4884fd2a7a31035c04a1084 : SUCCESS in 1h 15m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1e1e459d34c54f90a12e6717de5214ca : SUCCESS in 1h 56m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65188b733e534a7e85723240b55bf8af : SUCCESS in 53m 55s","accounts_in_message":[],"_revision_number":1},{"id":"38db677ca109241e05f0e6a9b43c9ed9b22f85e4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-14 16:46:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/864185 : FAILURE in 2h 06m 59s","accounts_in_message":[],"_revision_number":1},{"id":"e679fff44825eef230c3fb3f19dc12ccdd13c102","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-14 17:39:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/85/864185/1/check/iscsi-driver-alletra-6k/8d1e9a9/8d1e9a9.html : TIMED_OUT in 3h 01m 19s","accounts_in_message":[],"_revision_number":1},{"id":"d0f79b0ba5646f8d5c4de1fd428dd258c0aba123","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2022-11-15 01:47:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb03bfc05566549169d48036d80ae5351f66ed16","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-15 01:59:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 7m 30s","accounts_in_message":[],"_revision_number":1},{"id":"31a4be72a8a9867d3ac312f77c6b22e36154f446","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 02:06:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaae1f2020994b69b7a996c1cdbc7c40\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b1bb74906444abc97dcc4240f255e89 : SUCCESS in 10m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d4307a215bfb465c93e2f4087c5ef959 : SUCCESS in 14m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"094eb4bdbc629aea744fc90159e991068fe403a6","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-11-15 02:37:38.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/85/864185/1/20221115102321/NEC-ISCSI : FAILURE in 1h 13m 04s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/85/864185/1/20221115102321/NEC-FC : FAILURE in 1h 14m 14s","accounts_in_message":[],"_revision_number":1},{"id":"9b83eaecbcb5d648bc1b56970ebdb979db18275c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-15 03:50:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/85/864185/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/9b36a65/ : SUCCESS in 2h 00m 22s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/85/864185/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0b108dc/ : SUCCESS in 2h 02m 34s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/85/864185/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/883be12/ : FAILURE in 32m 32s","accounts_in_message":[],"_revision_number":1},{"id":"05ab4f3adadfeedf4669070f021b3769290730f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 04:23:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd1e60299768437c992dad5b5b619835\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/493faf5895384d818bbc71227023085b : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7e84cfeee93410eafb9d3ca0906cbf6 : SUCCESS in 8m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/514ac96190c2494cb4473fc1b3320700 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfd9b3fc656a4e6bbcb11463d0668fc6 : SUCCESS in 9m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/406db9b817fa41989dfd2eb66c199f4d : SUCCESS in 10m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/11bbb464cf094e6a8282e5b97c7ed751 : SUCCESS in 56m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d5d1d69c07a34a388ee9ccec901ed4dc : SUCCESS in 1h 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2caf54df7d7f4d6589b218e8b2488f47 : SUCCESS in 51m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cbabdca651c45f4ac1739b4135e4d73 : SUCCESS in 8m 07s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/685d2c45ec5f449aa212bada536fec43 : SUCCESS in 12m 04s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/72268e29ced3411084e9c363311bf8e7 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/220e379d29784c91b43cd718d6c51e95 : SUCCESS in 7m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f71859e0c3c14536a8e08f35b91b36ff : SUCCESS in 5m 59s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/84dc6d98af964e00a99e84a8a5fd6682 : SUCCESS in 37m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c2a872a757fd49ce86f8c2840ca21aa6 : FAILURE in 5m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9ffbab338b6a4ea5bbbb39e196a982a5 : FAILURE in 1h 16m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f2a64e196cbf4255909f90c15e99ef4d : FAILURE in 1h 23m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4504ca78e73342a889dcfe1e921df67e : SUCCESS in 1h 36m 31s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/47e5be651eb040709fe9ea13bf1c6873 : SUCCESS in 1h 12m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d261059f08e8410d95285a9164787d7f : SUCCESS in 1h 37m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ecebe35bb19c4637a26942cca8479daa : SUCCESS in 1h 32m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f733c668f8f5468a90103234510d4e34 : SUCCESS in 2h 31m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d0ab60f717245319d5fc3d77fdcb948 : SUCCESS in 1h 04m 30s","accounts_in_message":[],"_revision_number":1},{"id":"31872d5c93a973c2e70d5ca77eb5207e5b5f7dc1","author":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"date":"2022-11-15 10:48:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"078fcbb4531e8fb5dcfea122bdced0f5f76862b5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-15 11:01:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 7m 30s","accounts_in_message":[],"_revision_number":1},{"id":"af942c8386b58ced0efccc9461c7b8ab958ecf3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 11:09:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3cd4dbcf83e44feb15686e5d82cb406\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f67789d0b814a499ff7b7443c60c45b : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9c1e63b1491445f0a29779d18fccaaca : SUCCESS in 15m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"414d95560dcedf621170eb1101b230acf03ff044","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 12:44:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dcffba8000d84245afb2cbab4af145f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eabb923b1e04cb897a907ffc57d50ee : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a421e14abe04e85bf7bb7ae4578409c : SUCCESS in 10m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6bf657eef5143ea87cccbc1e3778b5b : SUCCESS in 9m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a79e3ecc63b45ca87951011ab724a13 : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef69ccc7acbd469ab2d4fab5b4e37d4a : SUCCESS in 10m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/9eb8b95a13c645f5bbd3ee889d62a9de : SUCCESS in 1h 00m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8fda3ace547b4e4aa9ab4b067c7dc40d : SUCCESS in 1h 07m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6242d85bfaef43f99e8f67edd02c29bf : SUCCESS in 54m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1bd2061acc44655829029fb96a80ac1 : SUCCESS in 7m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/13d78e21dfd040d1b3a3289a06dd4442 : SUCCESS in 11m 42s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1a542978a58c45108200b29dc7b287a6 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6eca20cd585a4de0b934aa33fcea5eb9 : SUCCESS in 6m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/24be8f5881174cacbb244cd28092bd13 : SUCCESS in 5m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2b39dbfe1ff54d789619b0d76696546e : SUCCESS in 40m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/de219fb95ff44667a137792ba53eb564 : FAILURE in 5m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d2192284592b4c6fb20ef0610f372556 : FAILURE in 1h 15m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4cce7a3ec64940438ac9ad4f29820371 : FAILURE in 1h 22m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3f62083fbf2e4d9bb3ac89ccf12c1de1 : SUCCESS in 1h 24m 07s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6761738f1556433aa45aa2ed1e45a72c : SUCCESS in 1h 08m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c6fbc40a327c451cba9601dc428a7782 : SUCCESS in 1h 23m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/219c85b5f7ea4d33992aff54b26ac6be : SUCCESS in 1h 21m 18s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/abb30507578645048159d7da4b0645db : SUCCESS in 1h 54m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54361e40493d4017b98e3cedd8cf4679 : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8c3e28f4800026ffe9c4566655a6dfedc544eb2f","author":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"date":"2022-11-15 17:37:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b0bc09896bb5a14e6480fde6c7f60c984fe3b58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 17:57:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e921a22d68e94cef8a19f4989d792d93\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed96edfd7cca4bafad661b21d78e665f : SUCCESS in 14m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/077d77125ae7445181d2239a7c79dc73 : SUCCESS in 15m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59749e582e1b4b6684228637a6fb8b8cc1784246","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-15 18:31:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 7m 29s","accounts_in_message":[],"_revision_number":1},{"id":"5171771cda9259c995dcada4335e7b9f4f4375f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 19:26:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7078fdb8bf7c4db1a8a58ea307f89fa4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/317d62a621004782b0bd423bf3d655b6 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff28b9c8770647619ee4b348ca1752e5 : SUCCESS in 9m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/529c3cddc0194939a288cdd8cc40339f : SUCCESS in 7m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e8e6c535c2f4adb8542303ef35da1d2 : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4962d9d3f244330b2414656f5de1271 : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/b931df50de3149869bf8ac0523ffa578 : SUCCESS in 50m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1f73689f19c04679aa2284b66824d24e : SUCCESS in 1h 01m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9e84af953ab40858fb30e8fc40af491 : SUCCESS in 37m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebb3862a4cfd48afac17c92adbe57189 : SUCCESS in 6m 56s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f4bb9132c313401fb9be7f902d9e77ff : SUCCESS in 10m 20s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f43628060ae7431ebcedf1d5207d194b : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d070d8f574a346f5b0abb2bc48512eee : SUCCESS in 4m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a39daf710c10457da307c563e59fd482 : SUCCESS in 5m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/65993050bfec4666b876d6d93fed4e1e : SUCCESS in 31m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a8c4c16165f84397a0c96abaab84e0cf : FAILURE in 5m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/96d8d2e5063542f798c3fa7c3f167661 : FAILURE in 1h 03m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/17bee9a961de4eafad1d0372795f6de0 : FAILURE in 1h 07m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/15ba91da01c543ba81180b4298ef5710 : SUCCESS in 1h 20m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/10122fbe001a48f5851ace39ae60182a : SUCCESS in 58m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1cd6bd05ff9c4ffc92e7b6ea9c4b4a15 : SUCCESS in 1h 21m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8a5d9a4354824245ab7dc00177af08fc : SUCCESS in 1h 14m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ad01491c388d43eb8ab72cd38dacbd73 : SUCCESS in 1h 45m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f9aa0be43b944a3baa4184703fb806c : SUCCESS in 45m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a69adbe1b69ec14395e27f05047b41a8c8be120c","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2022-11-16 01:38:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"13a65e34d8518437815b8d9a0c0195e5d8e87370","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-16 02:13:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 10m 01s","accounts_in_message":[],"_revision_number":1},{"id":"9349539cd62271651da4531c0bdc1b4c91e133ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 03:07:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4c8535cf7b144b48b7438e39c610cbc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/987ca884e57c4eb9bb88d86b8d4b1be7 : SUCCESS in 16m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bd6bc4567f4c4fb8b1c3c5cfb09a2b79 : SUCCESS in 36m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3f5ec02c3f2b02dce128d20d1fc6026d6f2082e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 03:42:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e73593f256d8480dbd51ed2df6094d8e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28296b2d877a49c08a085338ff12fe61 : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06e6d47f84304b4fa5efe7a57f97c7e1 : SUCCESS in 10m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d574869cbdf641c4ba9bc54ec32ed6ef : SUCCESS in 10m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08b5e8cedc084c23b7c65f3624a71175 : SUCCESS in 10m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aa4d28d2b084e069c3b367907830606 : SUCCESS in 12m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/45a166b4f5ba482b834d3e88f79a842f : SUCCESS in 55m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b8ab6b0a120d4f0ebf2a049cb867b105 : SUCCESS in 1h 11m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4737fe05ff314d8ba86f8b0e6b0a80a7 : SUCCESS in 47m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4cd842a51664de9803d5c65be6d32ee : SUCCESS in 7m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7b65f37893ae4a13aeffdc7d3c837735 : SUCCESS in 12m 01s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8273872ba0e74ed4a7cb589793c8f694 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/72593c944b23403da99843ec73c62ac5 : SUCCESS in 6m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/03644f09fe394b968582fdc3c190c254 : SUCCESS in 6m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0af8d7562c9e45859ec76bee9f921f21 : SUCCESS in 35m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/05a518d173c84a7b887fd6a4ddf33628 : FAILURE in 6m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/809e4e7317de4091b8dbc93639f2a489 : FAILURE in 1h 17m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c059a64aedfd4e438a47dd17c86b54d1 : FAILURE in 1h 18m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d41d90da4bae481287ec7c2b70d5310a : SUCCESS in 1h 33m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/43da630e16554fb4be4bd6d1cb4227c1 : SUCCESS in 1h 06m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a5cabdb34e744cca985f7ff5a05fb6d5 : SUCCESS in 1h 19m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/8279d97f7db04fd2aed23a97829e95e4 : SUCCESS in 1h 22m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e9ae5cc989a548e88088fd19b32bf818 : SUCCESS in 2h 01m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8d66f57af01453d85c3574a29c03174 : SUCCESS in 47m 46s","accounts_in_message":[],"_revision_number":1},{"id":"e5015a22200f29a5ab77d9e2e5bdb246d78b393b","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2022-11-16 05:58:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3abd89354cc8218c8ab687a716d658976fbe79eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"date":"2022-11-16 05:59:49.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"abb36bc6e273f72fdfcde4c87e0bf7b898cced0d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-16 06:00:01.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1733 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"31a60d6c8630bbc09655174bfe4023c389fde3fd","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-16 06:23:05.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-85-864185-2/ns5_nfs : FAILURE in 22m 46s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-85-864185-2/ns5_iscsi : FAILURE in 20m 34s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"c717bd29d5065e382961348eb753eb8a009100dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 06:25:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9362bcac62646799df451988277c438\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ded3221431544bb3b5423b9e397b1fc9 : SUCCESS in 14m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8af0b50d00924f419228c652a9b43720 : SUCCESS in 18m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9f44d4595628c0ad4d6e9b774558af40320c5e7d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-16 06:26:38.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/85/864185/2/check/iscsi-driver-alletra-6k/3de7e3a/3de7e3a.html : FAILURE in 24m 14s","accounts_in_message":[],"_revision_number":2},{"id":"bdfaef12fb4a0bacaf99ccea39987d86b548f57f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-16 06:30:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 7m 29s","accounts_in_message":[],"_revision_number":2},{"id":"b596d3f75d94aad13e4ab8c5d4d187fc037a5c53","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-11-16 07:00:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/85/864185/2/ : FAILURE in 51m 46s","accounts_in_message":[],"_revision_number":2},{"id":"c4d694bcbfbda4578b7731c966b8e1b7148f6c90","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-16 07:43:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/85/864185/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/85ddefb/ : SUCCESS in 1h 38m 31s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/85/864185/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7604645/ : SUCCESS in 1h 42m 16s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/85/864185/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/107493c/ : FAILURE in 41m 45s","accounts_in_message":[],"_revision_number":2},{"id":"66f1993e3d6c64936d9aa0eb523a262a5df59d93","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-16 07:49:21.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/864185 : FAILURE in 1h 46m 37s","accounts_in_message":[],"_revision_number":2},{"id":"806873dae2d1a906d3d31168649c9a9a818f4ca3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-16 08:03:42.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/85/864185/2/check/iscsi-driver-nimble/d3c09b2/d3c09b2.html : FAILURE in 1h 51m 57s","accounts_in_message":[],"_revision_number":2},{"id":"e7bcaae8e4c1123a544f7d7dcc4e5f326e585c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 08:04:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0becb8937b104fcf8bf051690610ad13\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b27accf811540428f26f286860530a7 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ac363967ad14e9483745feaf96b1ef0 : SUCCESS in 9m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ccca2d371474f90adc446dce678d337 : SUCCESS in 7m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec50c6b6d29e4165aeea4510ae596d28 : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7123ad951bc04068a459ae951000cb72 : SUCCESS in 9m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/555c1deab5044df3b9583c94c8f42b6d : SUCCESS in 45m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/354b6124c81849619cb616ab583987c2 : SUCCESS in 1h 11m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b607351a7ad347899cbff146ded4508a : SUCCESS in 38m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4eff3ea77a0e46989d171a7e8a1bbc46 : SUCCESS in 9m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/764765eb3d204e15bd20456ea4cca641 : SUCCESS in 10m 34s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9bcd22f383554adaa0c5fcc933c20f69 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/851fa49b83bc43cf88e534069be1f30a : SUCCESS in 5m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5ddec79f43924bba91f37aacb47a305e : SUCCESS in 5m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b9efe7b670fd4b58bba566aca6296a28 : SUCCESS in 33m 03s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/47c45a4d254c46f59a905c802da2e536 : FAILURE in 4m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7329c68ede7f4f7e9dbf19acea0b258a : FAILURE in 1h 23m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9bc7cf450ea94cd5a8607596b4cbc7a8 : FAILURE in 1h 20m 13s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0800d72315844a979d9ca78fcb23fcb3 : SUCCESS in 1h 39m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9989c2d53e5b4c369372d5f5d4eb6d1c : SUCCESS in 1h 04m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c121d03e84c2424285a028f9d6ebaa6f : SUCCESS in 1h 21m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2ff8c7c18f374569a8840f94446613aa : SUCCESS in 1h 31m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b823c439191347c5bfdce86225eaed88 : SUCCESS in 1h 52m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b8228421f194276aa1c7c4df7272f86 : SUCCESS in 50m 53s","accounts_in_message":[],"_revision_number":2},{"id":"6c7aeac587f937e038d7623edd7d46b43d6e107e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-16 08:05:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/864185 : FAILURE in 2h 00m 27s","accounts_in_message":[],"_revision_number":2},{"id":"6c39c217c59a6bb09d1f135fff09403af3d7fcca","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-11-16 08:29:01.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/85/864185/2/20221116161442/NEC-ISCSI : FAILURE in 1h 12m 40s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/85/864185/2/20221116161442/NEC-FC : FAILURE in 1h 14m 15s","accounts_in_message":[],"_revision_number":2},{"id":"6cd378203280a77f378a67892c9bf6c48033efa6","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2022-11-16 23:50:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"268891d63fe3e488dabb39de37620d13fb1af015","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2022-11-17 01:38:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dc6711c84b8f631e6c7618bc1a15dc5a511f8c54","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-11-17 16:12:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"99f2b09bbfcd96b2825f0a352e1ebd172b6616b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-11-17 22:15:00.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0ca334d03fbfd0a6b2c630e6b416ba82ef96935e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-17 22:15:10.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1745 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"a86e0243b480a3b42ecfd4c494503158cc35408f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-11-17 22:17:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"53c489503095d08c9c8c52b0410bb27bf7061805","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-11-17 22:26:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9abc69514a8eae33b909e32f5c7e0166833052e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 22:34:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/235a5f46ea924fbc9b68ca6007c34377\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/98965a7de2b64260a4ca177663f508bb : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/600121cfe0524f8fa36383eed48daa9e : SUCCESS in 14m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1e6f88a1413a724e23cde0020fed7a63385b4531","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-17 22:38:38.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-85-864185-3/ns5_nfs : FAILURE in 17m 38s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-85-864185-3/ns5_iscsi : FAILURE in 22m 21s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"1257e8248cb4c1d974a469bc69b115944d6e1083","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-17 23:31:53.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/85/864185/3/check/iscsi-driver-nimble/f3be331/f3be331.html : FAILURE in 48m 40s","accounts_in_message":[],"_revision_number":3},{"id":"ff4358b6a8a3b8cccc07354557ec943f29b7bc19","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-17 23:50:00.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/864185 : FAILURE in 1h 09m 58s","accounts_in_message":[],"_revision_number":3},{"id":"9804a27c732ff85ab26f56893e09ae402a7e13a6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-17 23:50:37.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/864185 : FAILURE in 1h 00m 36s","accounts_in_message":[],"_revision_number":3},{"id":"8359067537ccccdca479314989dac4a15adf26c4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-18 00:08:26.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/85/864185/3/check/iscsi-driver-alletra-6k/46d8763/46d8763.html : FAILURE in 1h 21m 18s","accounts_in_message":[],"_revision_number":3},{"id":"8b114e79fbdb2c582bb42b4252dc211621c3dd65","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-18 00:12:13.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/85/864185/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5c4d809/ : SUCCESS in 1h 56m 12s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/85/864185/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6dec15e/ : SUCCESS in 1h 55m 23s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/85/864185/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/fbd8631/ : FAILURE in 34m 26s","accounts_in_message":[],"_revision_number":3},{"id":"7b2630a9fa741a2db67eafefea5137f6afd8b79f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 00:24:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90d6ddbaa11045728543277f100edea2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf6b4feb8bf4420ab1c03ed3ecd7a929 : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f219638c31b447629d762632a14bc8df : SUCCESS in 9m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d8c98169b0c4377a27806212740c94a : SUCCESS in 9m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f67e1dbecf04f6c9f3751aa151d1a40 : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af542c37562f4feb910d219fee239ea8 : SUCCESS in 9m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/c98cc1be233f4b91a3cae577e609b89a : SUCCESS in 1h 01m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b741c2e581be4c029fa857d9f0e60c54 : SUCCESS in 1h 16m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/debde097de074fa7a98854c7d8377584 : SUCCESS in 51m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0c412fc5c26404aa9ca3c491764e369 : SUCCESS in 8m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a3306f7c8aa54f108c3feac674cda700 : SUCCESS in 13m 34s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/614cc36be6764c90859b61b46042daba : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/123eea97dbcb4dd8916dae668b2b10be : SUCCESS in 5m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ff8cd1c2e1424b0eb3646fd32df1bccd : SUCCESS in 5m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7d8f97b7c895499c972b3a9853525fbe : SUCCESS in 44m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5c8197aa8df143b79b029b7bdca1c8de : FAILURE in 5m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e3b0053774be4ecca113ebd52eb672a0 : SUCCESS in 1h 36m 57s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3558472c0dc544e7b8ae1a5e68d7c16b : FAILURE in 1h 42m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/59760822d16c4accbdea3ea4360c7116 : SUCCESS in 1h 58m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e767a801003e44ec8adc52b9e2f7da80 : SUCCESS in 1h 18m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5eb19a58a0e544d3ac9af1fa1a49d4da : SUCCESS in 1h 19m 24s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0afb6fdfa4744eefa4d2c8209b3692d1 : SUCCESS in 1h 32m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9743f3d36c45469498a94a2942a43687 : SUCCESS in 2h 04m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa223b01a0af430aadb234968086f7d1 : SUCCESS in 1h 01m 00s","accounts_in_message":[],"_revision_number":3},{"id":"d3a99b559832e38f4e15c4fe2af3a56dbed41415","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-11-18 00:39:46.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/85/864185/3/ : FAILURE in 53m 08s","accounts_in_message":[],"_revision_number":3},{"id":"96fb5e28b36317155bbb81469848342fd6b53027","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2022-11-18 03:02:40.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9d17f10399c21b1a15a78dfe46a953102e54a50c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-11-18 03:13:48.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/85/864185/3/check/DellEMC_PowerStore_iSCSI/f7b8e69/DellEMC_PowerStore_iSCSI/904 : ABORTED in 2h 26m 22s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/85/864185/3/check/DellEMC_PowerStore_FC/a5eac05/DellEMC_PowerStore_FC/1060 : ABORTED in 2h 24m 21s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"a41e29be976ad2568ef9ce15c01d80f6f5dcd325","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-11-18 18:34:56.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ad7e4ff3da6e54f9cb6b396b0b317a1529ec6c90","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-11-19 10:35:35.000000000","message":"Patch Set 3:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/85/864185/3/20221119182015/NEC-ISCSI : FAILURE in 1h 10m 43s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/85/864185/3/20221119182015/NEC-FC : FAILURE in 1h 15m 17s","accounts_in_message":[],"_revision_number":3},{"id":"ddd1b94c22f0a4835c94fc54485ab63fc6ac7988","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-11-21 06:36:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2761805fbe6d79175e01e973a4d7e8105f3b3c9a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-11-21 15:39:09.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"baf696dcc5079a10572de01c058e584e7a1aab15","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2022-11-21 19:42:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"70931281ea03d682db0327015a72f86749b69b46","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-11-21 19:48:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"612b9e2a8ea4bf6529ebc740e78d367556a46e8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 19:49:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ec152296d3a091d6ad0c3cc968723462fd738534","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 21:03:40.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/175fbaf1823949fc9a0f016a6222d2ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3551d83f5da74c2c93f7a53b5dfa4c5d : SUCCESS in 8m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a459e3530a814cc7b94e4cc2091fa058 : SUCCESS in 11m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec46b81afff54f59bfcea3aaaca32d9d : SUCCESS in 9m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c6bbdb0c32b495f83934ed76a822612 : SUCCESS in 11m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/846b4801e4e54a4ab44cd63b439ff264 : SUCCESS in 58m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/70ecfec8b6814790a743208a217393d8 : SUCCESS in 1h 13m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb8a0ec61ad64ce88ec9c42988897588 : SUCCESS in 48m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/825ff59495464b9ab18e9148f8cec8e3 : SUCCESS in 8m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/149163bd311148d2b35be9c6fa1b703c : SUCCESS in 1h 05m 41s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3e01c652316444438dc9d24c8959e2f3 : FAILURE in 1h 05m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bff13f8299b4562a41cc9c289c7ba3b : SUCCESS in 53m 45s","accounts_in_message":[],"_revision_number":3},{"id":"f0324804e15de46c580df09a8cc37c624f5c8397","author":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"date":"2022-11-22 01:21:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"29769a986ce2818eeef837013ba68c17a1fea82a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 01:47:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59157670dcfb45579496bed9c6324182\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a7a9da2881941a1a3577dc0568c9021 : SUCCESS in 11m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/499701c43f644db68d00f00aa3649720 : SUCCESS in 23m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8b5abdde0335ebda8825607c06b7b5815c90ec8c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-22 03:22:01.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/85/864185/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/eb5cf17/ : SUCCESS in 1h 58m 38s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/85/864185/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/eb488e6/ : SUCCESS in 1h 50m 36s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/85/864185/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/2f36d00/ : FAILURE in 33m 19s","accounts_in_message":[],"_revision_number":3},{"id":"99dd3661db7cd459d88a918ff9b5f4b27548f519","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 03:24:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9d8210b0eff4c60bb619ddfaf089659\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4d82ed7936e47b08476c485484b33a6 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/970e4cba98ad4b3db6d04d11a58e30ef : SUCCESS in 9m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b2f99a600f24ea997fe89e9c89c14fe : SUCCESS in 8m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f86f82f57ac45618dd82142ee7c9bf6 : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a70b21a324e4b76b0e6b251cb34d74e : SUCCESS in 9m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/b217c3ce02f74867a31df92094af0ef1 : SUCCESS in 1h 07m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/59b500966e8f44ca9d182da51e8b48d1 : SUCCESS in 1h 18m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24d18203f2384a95940523f28c576fe0 : SUCCESS in 56m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49314c5b2c064cb98ee816dc86d02e59 : SUCCESS in 8m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b73a27c52b97476994fc9e1617f9447e : SUCCESS in 14m 30s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9459459a3c0243918cbbc18b9d2ab6fd : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fea6a331afe24463929fa4add7aeb7ad : SUCCESS in 6m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/339185d5b90749b8a094b54cf4ba2e07 : SUCCESS in 5m 54s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e0f1c3b1ffb3425d9de532155b13feea : SUCCESS in 35m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2635527436f74ca7adf9a81540ac7e3e : FAILURE in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ac1fd18646734c409b17bd6d15ccf57f : FAILURE in 1h 04m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/60592052e5a64c26b70e679c259d259a : FAILURE in 1h 27m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7fdb0b5174d94bcda7f9fa61dcc7af82 : SUCCESS in 1h 25m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/98507e9788dd46f7925d345d44aa8641 : SUCCESS in 1h 06m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d3111efff23c41a39c906b9f205c8e1f : SUCCESS in 1h 10m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a7bc653793c148c282fd30ae873d0f7d : SUCCESS in 1h 22m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/83437da97a104245aa86cab9c4a1e125 : SUCCESS in 2h 00m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70bdc2d68f0d411898023ec0d5ac3a99 : SUCCESS in 51m 44s","accounts_in_message":[],"_revision_number":3},{"id":"fe2b78c23c209bcafe65ce87de5fda683727b94a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-22 04:17:40.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"8e7290838725f9388777aba2c45e34c537dbb4c3","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-11-22 04:44:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"adae231c7e2662b8c3fbd9b8bb1e6bbc91690f9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 05:14:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85cd5b0fefbe4e4f9cc31ec0781580e5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/13698c41f2464fdb92a1bbbb2da96c95 : SUCCESS in 14m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c4dc3db3eaac4a919a6ef950b9e7f080 : SUCCESS in 18m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bcafdcf86a53d4783d9adab3c18f3edc27aac778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 07:22:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f1d23c6f99e44e09346a68d97948b95\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e6da0bd08e249ec9f4a510a8daabfb7 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2226e7c59eb24a77a5ea10ef8f0f5236 : SUCCESS in 9m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2570331c353943b88a0c3b20f4d02544 : SUCCESS in 8m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/282fc10a9e664c07a6229a7af07fd87a : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83be68e668b341f09e9162f2310c3900 : SUCCESS in 10m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/9825431416a34c009f22af221a774ef7 : SUCCESS in 1h 09m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a15fb6d076ab435e809ca9f76b771315 : SUCCESS in 1h 23m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef6baa6cd5d046449059d50395beaa76 : SUCCESS in 59m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa61d3799f7742ae90e55f0bcfb22c97 : SUCCESS in 8m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e43a920f41bc416c951217755beb629a : SUCCESS in 12m 04s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/67b49f7c29e2471488f4fc1b3789c88f : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4ebaafd3440e48a78764fb0743929be1 : SUCCESS in 5m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/78a46b960acd4062b6dfdb9b445bb19d : SUCCESS in 5m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c5c556002ae74749a2234d5be1818a06 : SUCCESS in 48m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9836aa88e560465197ab06cff012424c : FAILURE in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/af3d5c325b6941638a9399b6cb9a56ea : SUCCESS in 1h 17m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/58ef07d94bfb492fb9383ad1e78dc74e : FAILURE in 1h 20m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f94e829c4879487f8457c48e3e32f3cf : SUCCESS in 1h 26m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/57cce3a7c3b44e11a227be4c26c0b55d : SUCCESS in 1h 07m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a77ed32b648b490caad37496d52e0583 : SUCCESS in 1h 15m 59s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ab96895273d642d49c3cf39ce3010416 : SUCCESS in 1h 20m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e976e7510fef40c88d31d71dbac107d6 : SUCCESS in 2h 36m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33152aa3c1ba4498901591be17745a0a : SUCCESS in 50m 36s","accounts_in_message":[],"_revision_number":3},{"id":"e6e639267b2f55e4c497b9e58e23a0d60bb2a02c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 07:22:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5cea3dd4e37dbed3c2a5df67c1fba8d9af04ad61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 08:45:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc55a2e7e25a4225afc11e25c225e12c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02706ee2a63c4cf682c17ef9a9b28030 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c91e015eeb4400381a98e337869ff0b : SUCCESS in 10m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51d36a01805d4924989455bcf1428e99 : SUCCESS in 9m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d98b6c837d64ac69c0bc1b37070ebee : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/d6e23dbc19d74d218eaf3d5f3d118e7e : SUCCESS in 49m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1e5c189f3b5b4911a8cafc8fb512bbd6 : SUCCESS in 1h 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12d71d8ba31d4bf9aa10f56b60cb9dfd : SUCCESS in 41m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/137e844d334d4ec2a201c4e838121f80 : SUCCESS in 7m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d4131cf462d0429e842e4824650df97b : SUCCESS in 58m 53s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0e071ae908a34280890bde838fc5b894 : SUCCESS in 1h 21m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c3f2eeb57b14a36b1c59b6f995f9f18 : SUCCESS in 52m 11s","accounts_in_message":[],"_revision_number":3},{"id":"daeca3ffd8e5aa4acb349ffe69baf3cd0e27ce84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 08:45:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"beada116558f525f686f38d6ee139387123c0b9a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 08:47:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c834e0853a51435cb0c23dcb8da05d4c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6d96fe204354c82b1c2bd182d29ac8c : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/202c715c808d46de8c79fc58c43bfb83 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"84dc4fb0599743ad19eac43ca2e10dedb5d5386b","revisions":{"a7419a24a6cac5e18671c65a8084af9fb91948f8":{"kind":"REWORK","_number":1,"created":"2022-11-14 14:35:18.000000000","uploader":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"ref":"refs/changes/85/864185/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/864185/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/864185/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/864185/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/864185/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/864185/1"}}},"commit":{"parents":[{"commit":"ebc27d371450dccbba49443ab9094484a4f9905c","subject":"NFS: Fix generic revert to snapshot flow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ebc27d371450dccbba49443ab9094484a4f9905c"}]}],"author":{"name":"Christophe Drevet-Droguet","email":"cdr@antemeta.fr","date":"2021-02-04 17:18:46.000000000","tz":60},"committer":{"name":"Manojkatari","email":"mkatari@redhat.com","date":"2022-11-14 14:35:18.000000000","tz":0},"subject":"[PURE] support IPv6 / add parameter pure_iscsi_cidr_list","message":"[PURE] support IPv6 / add parameter pure_iscsi_cidr_list\n\nPure storage: Add support for IPv6 in pure_iscsi_cidr parameter.\nPure storage: Add parameter pure_iscsi_cidr_list to support a list of networks.\n\nChange-Id: Ib04e49dac9fdabd81809ca389553ce3638bb9bbf\nCloses-Bug: 1910143\n(cherry picked from commit 7aa5e95d48c1bd0110001fb967c80e101f5b94dd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a7419a24a6cac5e18671c65a8084af9fb91948f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a7419a24a6cac5e18671c65a8084af9fb91948f8"}]},"branch":"refs/heads/stable/wallaby"},"5ca770fe0f5a2cd42f185a4b866fabf61e3cc6a2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-11-16 05:59:49.000000000","uploader":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"ref":"refs/changes/85/864185/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/864185/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/864185/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/864185/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/864185/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/864185/2"}}},"commit":{"parents":[{"commit":"ebc27d371450dccbba49443ab9094484a4f9905c","subject":"NFS: Fix generic revert to snapshot flow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ebc27d371450dccbba49443ab9094484a4f9905c"}]}],"author":{"name":"Christophe Drevet-Droguet","email":"cdr@antemeta.fr","date":"2021-02-04 17:18:46.000000000","tz":60},"committer":{"name":"Manojkatari","email":"mkatari@redhat.com","date":"2022-11-16 05:59:49.000000000","tz":0},"subject":"[PURE] support IPv6 / add parameter pure_iscsi_cidr_list","message":"[PURE] support IPv6 / add parameter pure_iscsi_cidr_list\n\nPure storage: Add support for IPv6 in pure_iscsi_cidr parameter.\nPure storage: Add parameter pure_iscsi_cidr_list to support a list of networks.\n\nDepends-On: Ib001e2b4d1347754c2b46730bc10d86e8cdab7ad\nChange-Id: Ib04e49dac9fdabd81809ca389553ce3638bb9bbf\nCloses-Bug: 1910143\n(cherry picked from commit 7aa5e95d48c1bd0110001fb967c80e101f5b94dd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ca770fe0f5a2cd42f185a4b866fabf61e3cc6a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5ca770fe0f5a2cd42f185a4b866fabf61e3cc6a2"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"},"84dc4fb0599743ad19eac43ca2e10dedb5d5386b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-11-17 22:15:00.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/85/864185/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/85/864185/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/864185/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/864185/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/85/864185/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/85/864185/3"}}},"commit":{"parents":[{"commit":"273b3394d36f21c5364a0140c350d0a77c44676d","subject":"[stable-only] Use os-brick from source in wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/273b3394d36f21c5364a0140c350d0a77c44676d"}]}],"author":{"name":"Christophe Drevet-Droguet","email":"cdr@antemeta.fr","date":"2021-02-04 17:18:46.000000000","tz":60},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-11-17 22:15:00.000000000","tz":0},"subject":"[PURE] support IPv6 / add parameter pure_iscsi_cidr_list","message":"[PURE] support IPv6 / add parameter pure_iscsi_cidr_list\n\nPure storage: Add support for IPv6 in pure_iscsi_cidr parameter.\nPure storage: Add parameter pure_iscsi_cidr_list to support a list of networks.\n\nDepends-On: Ib001e2b4d1347754c2b46730bc10d86e8cdab7ad\nChange-Id: Ib04e49dac9fdabd81809ca389553ce3638bb9bbf\nCloses-Bug: 1910143\n(cherry picked from commit 7aa5e95d48c1bd0110001fb967c80e101f5b94dd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/84dc4fb0599743ad19eac43ca2e10dedb5d5386b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/84dc4fb0599743ad19eac43ca2e10dedb5d5386b"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
