)]}'
{"id":"openstack%2Fcinder~864287","triplet_id":"openstack%2Fcinder~master~I5e065a3c16d438df604a169deca9fb1852f13d6b","project":"openstack/cinder","branch":"master","topic":"infinidat-add-local-volume-migration","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-01-28 19:22:57.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"35075":{"account":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"last_update":"2023-01-28 20:49:40.000000000","reason":"Change was submitted"},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2023-01-28 20:49:40.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-01-28 20:49:40.000000000","reason":"Change was submitted"},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2023-01-12 17:40:00.000000000","reason":"\u003cGERRIT_ACCOUNT_20813\u003e replied on the change","reason_account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}}},"hashtags":[],"change_id":"I5e065a3c16d438df604a169deca9fb1852f13d6b","subject":"Infinidat: add storage assisted volume migration","status":"MERGED","created":"2022-11-15 10:26:36.000000000","updated":"2023-01-29 00:09:40.000000000","submitted":"2023-01-28 20:49:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864287-infinidat-add-local-volume-migration","meta_rev_id":"6a257a53a3b1861903bbb7b3160e1076d82c15ba","_number":864287,"virtual_id_number":864287,"owner":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-28 20:49:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"all":[{"value":1,"date":"2022-12-20 14:11:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":2,"date":"2023-01-28 19:22:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-17 12:09:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":2,"date":"2023-01-12 17:40:00.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2023-01-28 19:22:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"reviewers":{"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-15 10:26:43.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-11-15 10:30:06.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-11-15 10:47:39.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-11-15 10:51:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-15 11:33:40.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-15 12:01:55.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-15 12:07:24.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-15 12:22:21.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-11-15 12:43:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-15 12:54:35.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-11-15 12:57:29.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-11-15 13:26:24.000000000","updated_by":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"reviewer":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"state":"CC"},{"updated":"2022-11-15 13:29:33.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-11-15 14:03:00.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-15 15:44:49.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2022-11-15 17:18:57.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-15 19:06:46.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-11-15 22:37:41.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-12-09 09:02:26.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-12-16 10:35:45.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2022-12-19 22:38:36.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2022-12-20 14:11:19.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2022-12-23 19:55:52.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2023-01-11 15:27:12.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2023-01-20 16:40:00.000000000","updated_by":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-01-21 11:46:07.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-21 16:31:18.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"c3976136a418950d014f94600b41c61b1ad4dec7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2022-11-15 10:26:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46054929bc6da59484fea2698a0ab3828b854898","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-15 10:26:43.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1729 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"bffee63911703423f9cc2e79a4056dce52394396","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-15 10:30:06.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"f85538441294fac811f7dec5ab3908481a7f0cab","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-15 10:47:39.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 7m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a66dcf30568f67f03aa1547e1075cce7456d015a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 10:51:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/543ee194f5164f7eb3563195e41de74d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eb2c8bc12d89410a80f865cbf59be109 : SUCCESS in 13m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ee1dce77607b438e9af90d8521722730 : SUCCESS in 17m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0eb4179de4881d3842a888e991bc83844784d467","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-15 11:23:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/87/864287/1/check/iscsi-driver-alletra-6k/5c13894/5c13894.html : FAILURE in 54m 09s","accounts_in_message":[],"_revision_number":1},{"id":"0a6501abbe58abb847f80d86276cc6070a2a572a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-11-15 11:33:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/87/864287/1/ : FAILURE in 57m 24s","accounts_in_message":[],"_revision_number":1},{"id":"08133885e2c2419abf6cc30eafb8496c16ca0fa6","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-11-15 12:01:55.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/87/864287/1/check/fujitsu-eternus-dx-iscsi/37c3cc0 : SUCCESS in 1h 31m 47s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/87/864287/1/check/fujitsu-eternus-dx-fc/d1f3138 : SUCCESS in 1h 34m 16s","accounts_in_message":[],"_revision_number":1},{"id":"d1c74b63425dd84b52fef6323e04b2df6ab19132","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-15 12:07:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/864287 : FAILURE in 1h 37m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9a0deaccf1742e69204430bb2c4db838e0772767","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-15 12:22:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/87/864287/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/5a49942/ : SUCCESS in 1h 28m 43s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/87/864287/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/82425d8/ : SUCCESS in 1h 54m 17s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/87/864287/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/acea484/ : SUCCESS in 1h 39m 47s","accounts_in_message":[],"_revision_number":1},{"id":"27e1830477a8cf5a8d30fc5758a8a2e84d96a5b3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-15 12:34:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/864287 : FAILURE in 2h 04m 35s","accounts_in_message":[],"_revision_number":1},{"id":"aa90d2840e7915ff9154dddcd38d8ed4db557801","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 12:43:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b3a9c9bd53540668492469945b4a974\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a02b381a906644ada5645346414c3d14 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76b356673ebd4c8187a3b1564a9a1b2b : SUCCESS in 10m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d4142bfdfb544ee9d238abe7a0fdcc7 : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b59a3196e468405bb34a9375f65799d2 : SUCCESS in 12m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/9ba7e39ee24d49378caa761e1e699496 : SUCCESS in 40m 00s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fa4becdb5b3c4209a4062a85540e3ef9 : SUCCESS in 53m 42s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/13aa61e1aefc4381b9974d59767a8a69 : SUCCESS in 1h 03m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85a50499eb574abd9bf94bca35272eff : SUCCESS in 42m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f37603cbef5f42f59c43f54d553663b7 : SUCCESS in 7m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9d503d2287b8448fb0de715697dfa6b3 : SUCCESS in 11m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/48a4c18699fe493c9142137919fc192c : SUCCESS in 5m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/09054103d2774007b8d18b9583ed388e : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/16dfb029050f4777bc9ca3b9860b97d8 : SUCCESS in 5m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fc7e2d3182f449de9ac7594c84a3c468 : SUCCESS in 5m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c40af0f3b48a415cb877a4e9e3e9a559 : SUCCESS in 33m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/377afd2279fc4d1191d8dd281add070d : FAILURE in 5m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/55c7cf2f3ba94c8b9468ae8a6c5e1aed : SUCCESS in 1h 12m 36s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/12c13ac11307415e8b1775cce9293031 : FAILURE in 1h 25m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f2c9a050976e4347b396dccdcee1a61f : SUCCESS in 1h 33m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/81c6cb4b59664c4ca1798ccae14f3209 : SUCCESS in 1h 07m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9157b26e2ef74cf38ea3416d104d6024 : SUCCESS in 1h 08m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9073d12ea98545afa98148ed7350d2e1 : SUCCESS in 27m 45s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4865ceddd0a643d39c59feb1f63be2c9 : SUCCESS in 1h 25m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/626d3af96a424237aedd884a148eb9ae : SUCCESS in 2h 16m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0114e1eb49014f69a173f42955a18eee : SUCCESS in 43m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1f29ecbe8ceb254b935f37f6b863ac09dd81433b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-11-15 12:54:35.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/15_11_2022_12_43_50\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"8c7d9c288b0a118fe896753023ff0e092c9ca6b3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-15 12:57:29.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/87/864287/1/check/DellEMC_Unity_FC/e264ba5/DellEMC_Unity_FC/1996 : SUCCESS in 58m 12s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/87/864287/1/check/DellEMC_Unity_iSCSI/7b5d451/DellEMC_Unity_iSCSI/1843 : FAILURE in 2h 30m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"e265a18441b6bd6ba6511b6e27de47bb911bfab9","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-11-15 13:26:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/864287/1/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/5d5ec42/ : FAILURE in 41m 13s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":1},{"id":"48a5d92fd0c372e9edb9ca82678c88441c1f6b13","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-11-15 13:29:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/864287/1/upstream-check/cinder-tempest-plugin-ontap-fcp/1b78d81/ : RETRY_LIMIT in 8m 13s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/864287/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/2ae5bd5/ : SUCCESS in 42m 30s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/864287/1/upstream-check/cinder-tempest-plugin-ontap-nfs/a9e3d51/ : SUCCESS in 44m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"01abd0bac26efdb4ec58ed48c437d58eac6d3b28","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-15 14:03:00.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-87-864287-1/ns5_nfs : SUCCESS in 3h 28m 22s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-87-864287-1/ns5_iscsi : SUCCESS in 3h 35m 54s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"835386829b44e8a465b502fd122e03a98a63aa55","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2022-11-15 15:44:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n* Infinidat-FC http://openstack-ci-logs.aws.infinidat.com/#logs/refs-changes-87-864287-1 /FC/ : SUCCESS in 1h 11m 8s\n* Infinidat-iSCSI http://openstack-ci-logs.aws.infinidat.com/#logs/refs-changes-87-864287-1 /iSCSI/ : SUCCESS in 1h 2m 47s\nFor more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"991929c749e6b7f1598f01f7fb05553fcb8f058d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-11-15 17:18:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/87/864287/1/20221115224714/NEC-ISCSI : SUCCESS in 3h 30m 57s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/87/864287/1/20221115224714/NEC-FC : SUCCESS in 3h 31m 39s","accounts_in_message":[],"_revision_number":1},{"id":"89eadfd68fb12441e87d445a378abc2e1003943a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-11-15 19:06:46.000000000","message":"Patch Set 1:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/824","accounts_in_message":[],"_revision_number":1},{"id":"8894f7504b1c0509b48000e7359c1e403eeef2a1","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-11-15 22:37:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/87/864287/1/20221116012419/NEC-V-ISCSI : SUCCESS in 2h 48m 14s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/87/864287/1/20221116012419/NEC-V-FC : SUCCESS in 6h 13m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f3f7193d57e45e87c126faa2250a62f44b66cfd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2022-11-24 20:49:54.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a66f19afba2f50338c22ee75f48c29332126094e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-24 20:50:08.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1793 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"09d899de4cf55c272905651fe5504d60bfcfc710","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-24 20:59:46.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"41f5324bf91aceb2e4b7c1cb49d4c1568b8ea75e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-24 21:01:27.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"0d500df4407738e28ccb5554ce38133e2183ad2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 21:17:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8de0cfa7ab16411197fd395c377b673e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/18d00f6b628d4159a719c4c7aca301b9 : SUCCESS in 17m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a4f670fa75c544418fc711ad5aaf47ff : SUCCESS in 18m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"106f7a06db24646ab8d2c2a970a34241f1e88347","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-24 21:54:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/87/864287/2/check/iscsi-driver-alletra-6k/2e90d3a/2e90d3a.html : FAILURE in 53m 12s","accounts_in_message":[],"_revision_number":2},{"id":"08152634bfc22437408bd6d1d65b8a152f89404a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 23:02:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e02208ca53974dee8b44c5f2643302ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4556d22f34944825b45aecf9506cc68b : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9064605b3a6547fe87e7e98c0fc2b6ec : SUCCESS in 9m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac5140ccd62447a68b8baf1999146e80 : SUCCESS in 9m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ca4a7c5dc6a4a819046020801370a9a : SUCCESS in 12m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/2138243c2ef64c459bd8eb7fbc7a5394 : SUCCESS in 45m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7b52c5576a454caa919c49581560bf77 : SUCCESS in 50m 23s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f86ada4d5f6e4af4b404b3d69f6119f9 : SUCCESS in 1h 15m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3607fd8c9d744db95752d2ade97a7c5 : SUCCESS in 51m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a790b41a05854044a992873205273559 : SUCCESS in 7m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cb24a08bc8a946c0bafcc39ed460a12a : SUCCESS in 14m 58s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4cb06b1025f64099b8544cc1aa79f152 : SUCCESS in 5m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a0c6915174f44e4e9bb6af87c7cc1dac : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/627eb684ffaf4fb989f8e45ea6b110f2 : SUCCESS in 5m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5de5fff2a47d46aca083275d11b7c6be : SUCCESS in 6m 30s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6af508ae48454f06ba74cab4bb039fd8 : SUCCESS in 54m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/966aeeb97bf94aeeb92500032a1d418a : FAILURE in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/85e400f6df29459c8e4770cef0830afc : SUCCESS in 1h 00m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4948fe04c8a04f7bb93093ee8a071872 : FAILURE in 1h 16m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c7ca079ee7df4faf86bd3776ba0f1aeb : SUCCESS in 1h 39m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/76d7d275ca0743b59aa248f08f87203f : SUCCESS in 1h 11m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/8e2e011f64c84e12abcaf34f50bb6b5c : SUCCESS in 1h 40m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cea781678bfd4fe7a656e359b6879366 : SUCCESS in 35m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e96973dde1bf49068a0008536c499a0f : SUCCESS in 1h 28m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1fe928443bed41e5a2c45e902a2484b0 : SUCCESS in 2h 09m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/615bab6433f64992ab35733d57953023 : SUCCESS in 49m 52s","accounts_in_message":[],"_revision_number":2},{"id":"cdbfd0fbf108fb5dcf3d544c81675695ed41c395","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-11-24 23:14:47.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/864287/2/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/d1e60cb/ : RETRY_LIMIT in 7m 50s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":2},{"id":"3167ce89b5846e2d71d530909d91451f600375aa","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-11-24 23:49:39.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/864287/2/upstream-check/cinder-tempest-plugin-ontap-fcp/bf93daa/ : RETRY_LIMIT in 8m 07s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/864287/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/b9bb5ff/ : SUCCESS in 45m 52s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/864287/2/upstream-check/cinder-tempest-plugin-ontap-nfs/78dbd21/ : SUCCESS in 43m 22s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"cc60e261e2ad22a1ea0157cc5d9b7bc26b2f2183","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-24 23:59:36.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/864287 : FAILURE in 56m 47s","accounts_in_message":[],"_revision_number":2},{"id":"f653aee4f26ee65a18a40f7dfd869ffec0a70d42","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-25 00:28:47.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/864287 : FAILURE in 45m 44s","accounts_in_message":[],"_revision_number":2},{"id":"1eb87a55b8f4ee517a9a56094fa35963c234d5b9","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-11-25 01:09:53.000000000","message":"Patch Set 2:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/87/864287/2/check/fujitsu-eternus-dx-iscsi/bf8f429 : SUCCESS in 1h 37m 54s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/87/864287/2/check/fujitsu-eternus-dx-fc/aa84073 : SUCCESS in 1h 48m 49s","accounts_in_message":[],"_revision_number":2},{"id":"8dd5ff95c08e7b013925b64b0a4dcaad9a1b341a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-25 01:35:01.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-87-864287-2/ns5_nfs : SUCCESS in 3h 30m 18s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-87-864287-2/ns5_iscsi : SUCCESS in 3h 36m 50s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"0c1995931c2ca86a29f59631bde7d873c28bc1eb","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-25 05:11:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/87/864287/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/8f781f9/ : SUCCESS in 1h 35m 16s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/87/864287/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6ec02e6/ : SUCCESS in 1h 34m 47s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/87/864287/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/af98576/ : SUCCESS in 1h 59m 57s","accounts_in_message":[],"_revision_number":2},{"id":"5286ac9d5bed61aa2407ea1af5f5e6d9eb501cd2","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-11-25 07:48:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/87/864287/2/ : SUCCESS in 1h 40m 36s","accounts_in_message":[],"_revision_number":2},{"id":"3983792120ecf97f9f5ea96299fe4249e24110ef","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2022-11-25 12:56:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n* Infinidat-FC http://openstack-ci-logs.aws.infinidat.com/#logs/refs-changes-87-864287-2/FC/ : SUCCESS in 1h 8m 45s\n* Infinidat-iSCSI http://openstack-ci-logs.aws.infinidat.com/#logs/refs-changes-87-864287-2/iSCSI/ : SUCCESS in 1h 45m 59s\nFor more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"63cb7e8c92ef2a4d83ab4fbbee068e535ef24360","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-11-27 07:18:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/87/864287/2/20221127124603/NEC-ISCSI : SUCCESS in 3h 29m 12s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/87/864287/2/20221127124603/NEC-FC : SUCCESS in 3h 32m 48s","accounts_in_message":[],"_revision_number":2},{"id":"78827f6204c3fb1baae96f7e2661b2e40e92c271","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-11-28 21:12:24.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/87/864287/2/20221128234046/NEC-V-ISCSI : SUCCESS in 5h 48m 20s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/87/864287/2/20221128234046/NEC-V-FC : FAILURE in 6h 31m 35s","accounts_in_message":[],"_revision_number":2},{"id":"ec159230271416b6cf4acd3f13a1fcc0d13a7556","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-12-09 09:02:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"722430b91acb445c88b4bc1952068115bbdd27c7","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-12-16 10:35:45.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0ab24a5e0d5abb6a0f1e9e633a6bb0c5c4153ee1","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-12-16 15:11:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"90439469be6a8d8746a450aea543f9f969510a1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2022-12-19 13:46:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"599623879d529810a7b4835c126c7a70673cc84f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-12-19 13:47:06.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1872 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"bb48c3ee936391737247e52fd8a5a69354c8e49d","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2022-12-19 13:48:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fb9c0e9d15c30f99077cc1ff6cd38b95aa4e70e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 14:13:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/323ce63fae3a4c9b835de297788e7df4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e2727831f084f69b2dda232f101b738 : SUCCESS in 22m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8089b403f3a540c8942c999b2e0766fb : SUCCESS in 20m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a689a25d6d3ca84b5aa0ce07aecad3fc1ab3aa1d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-19 14:24:24.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/87/864287/3/check/iscsi-driver-alletra-6k/02a4c3f/02a4c3f.html : FAILURE in 32m 37s","accounts_in_message":[],"_revision_number":3},{"id":"8c9e65a06b5ce1d9d166e7d1881ad048083ea55d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-19 14:47:13.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/864287 : FAILURE in 58m 38s","accounts_in_message":[],"_revision_number":3},{"id":"7361ff1170a0f25829cbedbaeeb3f17a2171e893","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-19 14:55:08.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/864287 : FAILURE in 1h 06m 30s","accounts_in_message":[],"_revision_number":3},{"id":"619f866deb355b399c8f48fc6c882aba716bdec3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2022-12-19 15:18:55.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n* Infinidat-FC http://openstack-ci-logs.aws.infinidat.com/#logs/refs-changes-87-864287-3/FC/ : SUCCESS in 1h 3m 32s\n* Infinidat-iSCSI http://openstack-ci-logs.aws.infinidat.com/#logs/refs-changes-87-864287-3/iSCSI/ : SUCCESS in 1h 15m 55s\nFor more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"1e9e17e26bc8d9eb2c469d5daf4d266c3386079c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-19 15:20:49.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : MERGER_FAILURE in 4m 48s","accounts_in_message":[],"_revision_number":3},{"id":"4074c890aaa4cfabe041f0df61fdbbace8b9abf6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-12-19 15:40:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/87/864287/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/f0d0a46/ : SUCCESS in 1h 39m 53s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/87/864287/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/070368f/ : SUCCESS in 1h 43m 23s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/87/864287/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/7021cbb/ : SUCCESS in 1h 52m 16s","accounts_in_message":[],"_revision_number":3},{"id":"40f9f4b4e819ba7e52c56229cbc429d820a4f56c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 15:56:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e29f9285af7b4937b33a06c3658ca806\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5dab2cd8eab4e2e8b8a6c717725d63e : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74da8bb635204ac781ff68254f446758 : SUCCESS in 10m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/177783cd17534d72a3f4035fdd52451f : SUCCESS in 11m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5fbcb0a43d2496c8e7b66b25b154c7f : SUCCESS in 13m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/058009afbd3a4cb49accf33a52904c48 : SUCCESS in 1h 03m 37s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0334360f5a4645c7bcbee44678ea5a5e : SUCCESS in 1h 02m 23s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/87eba2ec959c4d21aee7a274b7b38814 : SUCCESS in 1h 36m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2a79cb082564b7bad615a9e2592dab3 : SUCCESS in 58m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d65b6eaeb9bb49f590fbf9c01c107ab8 : SUCCESS in 10m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3ce4d5dc93c44c1fa859300d4edd9a81 : SUCCESS in 18m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d432604ef09e4170a5561f47dd45bc13 : SUCCESS in 6m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3a8f399af8d144ac913113318b1ccd55 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1b41612aa9ca4ceb83cc5ca5063ce47f : SUCCESS in 6m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1c96ec2ee38c480b81e6a7142f40a022 : SUCCESS in 7m 01s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/83bffa3dc0c4461f8655b817c5aa927e : SUCCESS in 57m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a5f5df796ec64dd9a707784c12147d6c : FAILURE in 6m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ca51089874904e6bacc98a65e74c0a20 : SUCCESS in 1h 42m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/49fcc4480136472eac77a7ee60a65fc0 : FAILURE in 1h 33m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b25cb7ab66c84f7faebbd78972692fbb : FAILURE in 2h 01m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0bc72a8b67464eff8fb3f8d384c028db : SUCCESS in 1h 23m 02s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/02f60db672ec4773b3f011d1c801af8d : SUCCESS in 1h 02m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/46f954a3aed6467e847926092cdb670a : SUCCESS in 38m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/61589bda5e094a37a68ff7df12f7d5a1 : SUCCESS in 1h 26m 24s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/34e89f4e1e434f029ba8b46ab762c80f : SUCCESS in 2h 01m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e9ce2e263394fa0b2cb1d908d7d2719 : FAILURE in 18m 10s","accounts_in_message":[],"_revision_number":3},{"id":"d83144323107489d45ba4369e1158605025ef818","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2022-12-19 15:57:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cb026ae5285001fa1c459eb707985853c79bfdd5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-12-19 16:05:47.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"fbccee8d23500f49d7f34ee5512e8723ec9e36be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 16:18:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78165c93058346de96d553c9b9ba46fb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d8541979c9042adb4c7c2e6ce6f0ef6 : SUCCESS in 18m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/23aa96f520f64fad8ee7da783ecc7dfe : SUCCESS in 18m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9e72e5fed6c07f36bedc87eee14d3c2f845ba556","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-12-19 17:22:50.000000000","message":"Patch Set 3:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/885","accounts_in_message":[],"_revision_number":3},{"id":"1fee59435e70d9ef3acf61742c7d5db725b2f580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 17:53:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d49f227b38a42ec9b805dac49f12f91\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b733345d14bc4cedbd9fadb7c3aaddff : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/680285db50624a2b861622435ac84bcb : SUCCESS in 10m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7ae0c0bcdf54265bdc7c40e2dcbf303 : SUCCESS in 11m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa31e9ad8a2e4846a62b8a155ec50f08 : SUCCESS in 12m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/1252ea59852e4a3e88a7a3e9ed48631f : SUCCESS in 1h 04m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4614489008c84ad090bcb70f18b7c3fe : SUCCESS in 1h 04m 45s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/91921fc22db24ead89c391d911af5589 : SUCCESS in 1h 23m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b8cccc22f4d4d72a9e2969cf296a817 : SUCCESS in 1h 10m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d25abd4765249a5aaedac79a7917c55 : SUCCESS in 7m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/94ccb0aeb424474ba7bac3e68594cf99 : SUCCESS in 17m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bf4d85eec2b648708ca457d24c9651bf : SUCCESS in 5m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/da0a936730b046ba98eb6cbd677d0e6b : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a5815fff2bec454b88bfee8cc1b17924 : SUCCESS in 5m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8de86876c3254e8c8b16d383ce05df66 : SUCCESS in 5m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5db7012c91d3493092478d4325c3a733 : SUCCESS in 59m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cb351fdc306745a98d41e50357d615b2 : FAILURE in 5m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/241cbb63d3f9460b8646bd864d7bf8b9 : SUCCESS in 1h 25m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a68335328601470b810c2a5246db97b5 : FAILURE in 1h 19m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dfcc674681524db78154845a4f50f2cc : SUCCESS in 1h 50m 52s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/192fd91e00564b988a903573a375ac59 : SUCCESS in 1h 07m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/01a6c314ef8d48d4aae1386b81191d94 : SUCCESS in 1h 29m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ae7a5c83f2d74726bb006e000ad308e3 : SUCCESS in 31m 06s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b711a85bb5044a9fa6e3241f3cadaf98 : SUCCESS in 1h 14m 58s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f1ef662610c54ef594fe27cf2a581a26 : SUCCESS in 1h 53m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d96cbba635064a7b9babc319fcda043d : SUCCESS in 53m 28s","accounts_in_message":[],"_revision_number":3},{"id":"ee403c5b2b95518ce2e73b11ac02572a179b7507","author":{"_account_id":26077,"name":"NetApp SolidFire CI","email":"ng-openstack-solidfire-ci@netapp.com","username":"netapp-solidfire-ci"},"date":"2022-12-19 18:43:52.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-solidfire-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/864287/3/upstream-solidfire-check/cinder-tempest-plugin-solidfire-iscsi/aab8035/ : FAILURE in 48m 31s\n\nComment \"run-NetApp Solidfire CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_SolidFire_CI","accounts_in_message":[],"_revision_number":3},{"id":"0bf9f1ccecdc8a96813bbf0dbc5192c86a36cc2e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-12-19 19:07:18.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/864287/3/upstream-check/cinder-tempest-plugin-ontap-fcp/9f13802/ : SUCCESS in 1h 11m 49s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/864287/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/d997f22/ : FAILURE in 49m 07s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/864287/3/upstream-check/cinder-tempest-plugin-ontap-nfs/9650944/ : FAILURE in 48m 45s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"1891f884031ffee59c0a763dd89509da2fff388f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-12-19 19:41:59.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-87-864287-3/ns5_nfs : SUCCESS in 3h 26m 43s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-87-864287-3/ns5_iscsi : SUCCESS in 3h 30m 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"5dbf817abb34d4b23137808dc88555d715d98456","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2022-12-19 22:38:36.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/87/864287/3/check/DellEMC_PowerStore_iSCSI/dea1428/DellEMC_PowerStore_iSCSI/944 : SUCCESS in 1h 16m 19s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/87/864287/3/check/DellEMC_PowerStore_FC/fa56e3a/DellEMC_PowerStore_FC/1080 : SUCCESS in 1h 28m 06s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/87/864287/3/check/DellEMC_PowerStore_NFS/1c64737/DellEMC_PowerStore_NFS/580 : SUCCESS in 1h 23m 06s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/87/864287/3/check/DellEMC_PowerStore_NVMeTCP/4cf8d87/DellEMC_PowerStore_NVMeTCP/449 : SUCCESS in 1h 27m 24s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"2e0035e0d085696cb5713ef0b5899079198b4090","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-12-20 01:21:53.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/87/864287/3/20221220064434/NEC-FC : SUCCESS in 3h 34m 56s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/87/864287/3/20221220064434/NEC-ISCSI : SUCCESS in 3h 37m 15s","accounts_in_message":[],"_revision_number":3},{"id":"194b7cf9fe14f7c2fcc5676f5536c75fe91d7991","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2022-12-20 14:11:19.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f1d6604947f6429e21ba50c2eceefccaa48d55e5","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-12-21 02:29:31.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/87/864287/3/20221221042843/NEC-V-FC : SUCCESS in 5h 23m 27s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/87/864287/3/20221221042843/NEC-V-ISCSI : SUCCESS in 7h 00m 45s","accounts_in_message":[],"_revision_number":3},{"id":"e9aec1fd0382cc4ef6abb04fdc2959072d81fa94","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-12-23 19:55:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"68f98595d7da252541f35243413734f68f01c90b","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2022-12-23 20:10:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9d89084d19f404446f55ada07acd63982526afb6","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-01-12 17:40:00.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9ab0028656f864b1d2ea1fb25445c59f5591eaf1","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2023-01-17 12:09:15.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"04aaaf8229ce4432718329459aca7909a8c20e40","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-01-21 10:30:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8a10d74d1b313b9609b75d451792f8a609ee0f3f","tag":"autogenerated:zuul:check","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-21 10:31:22.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"9755be779413a172389476890677c7288614d3bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-21 10:59:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7cd6517bc9847f3b9f95f736d693a0d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a6f0ca62dd441ef9f087d85601d68b5 : SUCCESS in 25m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2c7f36335a6044e3b6279c138800f67f : SUCCESS in 23m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"75ef586b911a301269c4d17c5bccb509e79a35e4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-21 11:46:07.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/40e259ed84b94483b534d20d99f3871d : SUCCESS in 1h 13m 08s","accounts_in_message":[],"_revision_number":3},{"id":"d8024d424ee6273cc964959d0ccedf643972cf40","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-21 11:59:34.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/864287/3/upstream-check/cinder-tempest-plugin-ontap-fcp/82b9dcc/ : SUCCESS in 59m 20s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/864287/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/0ca3a8f/ : SUCCESS in 44m 01s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/864287/3/upstream-check/cinder-tempest-plugin-ontap-nfs/904cba4/ : SUCCESS in 43m 55s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/864287/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/80e40b8/ : FAILURE in 31m 45s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"5f4a7c671b7135e342d1f70031e6dfcd32722340","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-21 13:23:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e29f5e5a4184dcb8f270b8ba4343441\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f419ed621ec4474af95c1e4c8b2b071 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/336a67d9f0fe4f639caf60a52cec8445 : SUCCESS in 10m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3b50faefa904031a0947bdee5762d05 : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6bbd459c9314bc3bb87234aa2c2f421 : SUCCESS in 12m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/4e819f1ddee6445485c6d0712346fe00 : SUCCESS in 43m 50s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c3c3b2d4b30d4b46a9144d0e201ba53f : SUCCESS in 50m 22s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/81d1189796b8403e83e883799abc3d0e : SUCCESS in 1h 11m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/373ab032021046b89c3f91f9cb9f28eb : SUCCESS in 44m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04c4f715fc9f4e08aac35895a98e67f6 : SUCCESS in 6m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/29a7a9d8be984b84ace8f1d335fd6015 : SUCCESS in 13m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/58099a9a3b0643f6bcabea1b47670120 : SUCCESS in 5m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/22e62a0e03c44b718c084912e4c547b4 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8db9fc614c8c4a0189c0db5efb4fcd15 : SUCCESS in 5m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7caffc86feb8494a8a1289dbebdc78ac : SUCCESS in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cb490409b5ea426aa514f5c761421ee3 : SUCCESS in 34m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/111493326d1743e6882a982b862829c6 : FAILURE in 5m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e85d22d6851d42c89ec9eef09f5f64fe : SUCCESS in 1h 44m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/84fe09f04c52491ba96971b7a76741f8 : FAILURE in 1h 30m 11s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/367fb4fa83674dc58e51ad00c5d1de50 : SUCCESS in 2h 14m 04s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a784e255429d4ca3861345c55177fd06 : SUCCESS in 1h 15m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/846ca005a5cf4d02ab64174d219a498d : SUCCESS in 1h 20m 08s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8d10a28cfa7a4bfca0f596220648e508 : SUCCESS in 45m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9bbab0a800c24803a5b517c5bcf4a1f8 : SUCCESS in 1h 36m 04s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/118f02f46e104d07842f0f355928e40e : SUCCESS in 2h 17m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2a2185442c448c9b029bf94eda6daf6 : SUCCESS in 1h 02m 14s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d13676efb1eb4614be2f18bcfd2e14d7 : SUCCESS in 1h 23m 47s","accounts_in_message":[],"_revision_number":3},{"id":"70eff94b6ba653e24576175756a5b132ddc13cee","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-21 14:24:39.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/864287/3/upstream-check/cinder-tempest-plugin-ontap-fcp/9b83cda/ : SUCCESS in 1h 00m 04s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/864287/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/f4cb6b2/ : SUCCESS in 43m 29s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/864287/3/upstream-check/cinder-tempest-plugin-ontap-nfs/6aad474/ : SUCCESS in 41m 58s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/864287/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/eee2d3c/ : FAILURE in 31m 07s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"379fb70ae4fee68deb296f5d45e2dd96e2fc08e6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-21 16:31:18.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/092b104fa05b4525a8d200bc71ba7508 : POST_FAILURE in 1h 13m 10s","accounts_in_message":[],"_revision_number":3},{"id":"43da20a701825fdcabde4511d13db510a659bbaf","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-01-21 22:16:10.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n* Infinidat-FC http://openstack-ci-logs.aws.infinidat.com/#logs/refs-changes-87-864287-3/FC/ : SUCCESS in 1h 8m 30s\n* Infinidat-iSCSI http://openstack-ci-logs.aws.infinidat.com/#logs/refs-changes-87-864287-3/iSCSI/ : SUCCESS in 1h 10m 13s\nFor more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"8cd39233c89eef3296c3b126d3f7b1da0c079632","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-01-28 19:22:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"da3d342094db5c2b56a10bf7cd69dd683c172f25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-28 19:23:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fca9979fe9e9316259f4d2d8cb626a657177d03b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-28 20:49:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d83cd398c074386a92f95b57283b7ff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95bdf245633e4f9f89c0708e88467e33 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7546cbca6bb41b4b156e6744db0afbe : SUCCESS in 9m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f937f3650034816acbb8d3f99c72ed2 : SUCCESS in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b90498a69414ccabd8dc3776b8bf87f : SUCCESS in 11m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/b0b1f65f8d524c639e19f81b7f590245 : SUCCESS in 46m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d3636f828b8741b1bfe696570a8a910b : SUCCESS in 1h 03m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecc912b60ea84322a9958cf7029435a5 : SUCCESS in 43m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f269446e8fb4dea8ae16c8b9ceabb8a : SUCCESS in 7m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1c48c7fa61064504a28125504d211697 : SUCCESS in 50m 56s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/56ff6370fae8470db20b0a6635951b45 : SUCCESS in 1h 20m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bbe145e635744f581c392bf5ad46325 : SUCCESS in 57m 20s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7fec5c45e944b94b758da5f7c5d711c : SUCCESS in 1h 01m 53s","accounts_in_message":[],"_revision_number":3},{"id":"299a339bca4bbc84a8d1a5b107dd588e9e0a1025","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-28 20:49:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9108be624fb9fb527d5ebdcffac2a37028c7128d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-28 20:50:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95d6f7f40dee466f9cd943b5dc27577a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37ae1a3ad2134efe9de2722829affff9 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79eb3fc2b9d247e6b9d8dd3fbcf98bbd : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3},{"id":"6a257a53a3b1861903bbb7b3160e1076d82c15ba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-01-29 00:09:40.000000000","message":"Patch Set 3:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/1020 \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7d442e5b5bf990d54385dfb43fe6de5ac5625b7f","revisions":{"2168a473df0831e3f251a6d2cdf1e2ef4e946d3a":{"kind":"REWORK","_number":1,"created":"2022-11-15 10:26:36.000000000","uploader":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"ref":"refs/changes/87/864287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/87/864287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/864287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/864287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/864287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/87/864287/1"}}},"commit":{"parents":[{"commit":"8154b08ea7a3ccb2fce083da7bd25ace8d53faee","subject":"Fix Infinidat driver consistency groups feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8154b08ea7a3ccb2fce083da7bd25ace8d53faee"}]}],"author":{"name":"Alexander Deiter","email":"adeiter@infinidat.com","date":"2022-11-15 08:51:19.000000000","tz":0},"committer":{"name":"Alexander Deiter","email":"adeiter@infinidat.com","date":"2022-11-15 08:56:13.000000000","tz":0},"subject":"Infinidat: add storage assisted volume migration","message":"Infinidat: add storage assisted volume migration\n\nAdding support for storage assisted volume migration\nin Infinidat driver. This patch implements:\n* migrate_volume()\n* unit tests to cover new code\n\nSigned-off-by: Alexander Deiter \u003cadeiter@infinidat.com\u003e\nChange-Id: I5e065a3c16d438df604a169deca9fb1852f13d6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2168a473df0831e3f251a6d2cdf1e2ef4e946d3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2168a473df0831e3f251a6d2cdf1e2ef4e946d3a"}]},"branch":"refs/heads/master"},"2c0cde10780bef11db3d3a115fd7d449dc5b9f8f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-11-24 20:49:54.000000000","uploader":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"ref":"refs/changes/87/864287/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/87/864287/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/864287/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/864287/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/864287/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/87/864287/2"}}},"commit":{"parents":[{"commit":"4f93938b8360c035fa2cef65010ffabaeff62944","subject":"Fix Infinidat driver consistency groups feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4f93938b8360c035fa2cef65010ffabaeff62944"}]}],"author":{"name":"Alexander Deiter","email":"adeiter@infinidat.com","date":"2022-11-15 08:51:19.000000000","tz":0},"committer":{"name":"Alexander Deiter","email":"adeiter@infinidat.com","date":"2022-11-24 20:39:34.000000000","tz":0},"subject":"Infinidat: add storage assisted volume migration","message":"Infinidat: add storage assisted volume migration\n\nAdding support for storage assisted volume migration\nin Infinidat driver. This patch implements:\n* migrate_volume()\n* unit tests to cover new code\n\nSigned-off-by: Alexander Deiter \u003cadeiter@infinidat.com\u003e\nChange-Id: I5e065a3c16d438df604a169deca9fb1852f13d6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c0cde10780bef11db3d3a115fd7d449dc5b9f8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2c0cde10780bef11db3d3a115fd7d449dc5b9f8f"}]},"branch":"refs/heads/master"},"7d442e5b5bf990d54385dfb43fe6de5ac5625b7f":{"kind":"REWORK","_number":3,"created":"2022-12-19 13:46:56.000000000","uploader":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"ref":"refs/changes/87/864287/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/87/864287/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/864287/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/864287/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/87/864287/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/87/864287/3"}}},"commit":{"parents":[{"commit":"d0d63dc7d8055b4719a6fd4c16f29d1d157be9d3","subject":"Fix Infinidat driver consistency groups feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d0d63dc7d8055b4719a6fd4c16f29d1d157be9d3"}]}],"author":{"name":"Alexander Deiter","email":"adeiter@infinidat.com","date":"2022-11-15 08:51:19.000000000","tz":0},"committer":{"name":"Alexander Deiter","email":"adeiter@infinidat.com","date":"2022-12-19 13:41:34.000000000","tz":0},"subject":"Infinidat: add storage assisted volume migration","message":"Infinidat: add storage assisted volume migration\n\nAdding support for storage assisted volume migration\nin Infinidat driver. This patch implements:\n* migrate_volume()\n* unit tests to cover new code\n\nSigned-off-by: Alexander Deiter \u003cadeiter@infinidat.com\u003e\nChange-Id: I5e065a3c16d438df604a169deca9fb1852f13d6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d442e5b5bf990d54385dfb43fe6de5ac5625b7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d442e5b5bf990d54385dfb43fe6de5ac5625b7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
