)]}'
{"id":"openstack%2Fcinder~865164","triplet_id":"openstack%2Fcinder~stable%2Fvictoria~Ib04e49dac9fdabd81809ca389553ce3638bb9bbf","project":"openstack/cinder","branch":"stable/victoria","topic":"pure_iscsi_cidrs-stable/wallaby-stable/victoria","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-12-12 18:41:51.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"34598":{"account":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"last_update":"2022-12-12 22:06:01.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-12-12 06:43:19.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"Ib04e49dac9fdabd81809ca389553ce3638bb9bbf","subject":"[PURE] support IPv6 / add parameter pure_iscsi_cidr_list","status":"MERGED","created":"2022-11-22 15:56:31.000000000","updated":"2022-12-12 22:07:25.000000000","submitted":"2022-12-12 22:06:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865164-pure_iscsi_cidrs-stable/wallaby-stable/victoria","cherry_pick_of_change":864185,"cherry_pick_of_patch_set":3,"meta_rev_id":"50905aafbd4d2034956ed20015ca50917c382692","_number":865164,"virtual_id_number":865164,"owner":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-12 22:06:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2022-12-12 18:41:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-11-28 07:38:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-12 06:43:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2022-12-12 18:41:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33082,"name":"Christophe Drevet","display_name":"dr4Ke","email":"cdr@antemeta.fr","username":"dr4Ke"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-22 15:56:31.000000000","updated_by":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"reviewer":{"_account_id":33082,"name":"Christophe Drevet","display_name":"dr4Ke","email":"cdr@antemeta.fr","username":"dr4Ke"},"state":"CC"},{"updated":"2022-11-22 15:56:41.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-11-22 16:28:23.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-22 16:55:38.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-11-22 17:08:32.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-22 17:42:35.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-11-22 18:13:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-22 18:46:49.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-11-22 21:30:00.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-28 07:38:45.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2022-11-29 07:00:11.000000000","updated_by":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-11-29 07:00:11.000000000","updated_by":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"a613f907745b2151d6f942c91ea3f2e3b6249801","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"date":"2022-11-22 15:56:31.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"a47805d9f2f8edda0a5a04f8e35191ecfee64766","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-22 15:56:41.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1767 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"e261ccc4ff2d223942efb43914cb55c242d3d48b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-22 16:28:23.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-64-865164-1/ns5_nfs : FAILURE in 29m 18s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-64-865164-1/ns5_iscsi : FAILURE in 31m 18s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"cd514e4747be79e665941294a148af5d5ca9918e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-22 16:55:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/64/865164/1/check/iscsi-driver-nimble/bb63a16/bb63a16.html : FAILURE in 54m 08s","accounts_in_message":[],"_revision_number":1},{"id":"315e44417f746d4a4c67031cd521c91ae12cce41","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-22 17:08:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/865164 : FAILURE in 1h 08m 01s","accounts_in_message":[],"_revision_number":1},{"id":"1d8066c7736e6887083572f9033614ed86e2a0f0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-22 17:39:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/865164 : FAILURE in 1h 40m 03s","accounts_in_message":[],"_revision_number":1},{"id":"d14312f8a1c58101eb672e7afc0df73110ca50c7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-22 17:41:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/64/865164/1/check/iscsi-driver-alletra-6k/29651f1/29651f1.html : FAILURE in 1h 41m 54s","accounts_in_message":[],"_revision_number":1},{"id":"d107a5b4941b30b26c6ef55369226f3262659928","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-22 17:42:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/64/865164/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/369047a/ : FAILURE in 1h 37m 14s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/64/865164/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/c06597d/ : FAILURE in 1h 44m 28s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/64/865164/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/52399b9/ : FAILURE in 35m 57s","accounts_in_message":[],"_revision_number":1},{"id":"70474fdb435a2e74204a62e7737adf1478ac4b15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 18:13:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7dab6d0789e9423c9583c27f14cc6335\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/833f0a01b0ed4edc8f8b1d4328ec09c5 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bc20107579c4c5da19054eb534eb5c1 : SUCCESS in 8m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7979c92800f470caf5fa6b6869ef02c : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0faac9be1284eb2992148275237bf56 : SUCCESS in 10m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/395e83385009459789af7af4223eacd8 : SUCCESS in 41m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/71581aee3b1547ed9c5ffe840523e9c9 : SUCCESS in 1h 00m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/620796c9406b4c70807956fc8947229c : SUCCESS in 40m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e34565e37234f2f951f85f19459687e : SUCCESS in 6m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/94b62532ff634013b287f6f2ffa8be2d : SUCCESS in 12m 11s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/36f13ef1cafc4a9798c29b68a388e449 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bb2b45e5d81b42e293609e6cb04c2941 : SUCCESS in 5m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2feec9436f534f2f82df7ae6a6a99940 : SUCCESS in 6m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5c91bfca8e81421eabb081c1b348f5f9 : SUCCESS in 36m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4e7d3b553a4d4162ba62fa027e31c1bf : FAILURE in 6m 01s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8bb3595932df40c4af4af1fc82c94b97 : SUCCESS in 1h 08m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b2e63e8c49c644bcaf27e8b2ae6cad7b : FAILURE in 1h 39m 41s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/664d1eb2266b46278812be838083a509 : SUCCESS in 1h 26m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7cdf0d76d3304b6e8b0005f6e7f3f7d4 : FAILURE in 1h 09m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6ec84f46a6a3499ba2bf284775ec9a3e : SUCCESS in 1h 08m 22s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6e2ad65b177846d79d0dfd2583ab9ce2 : SUCCESS in 1h 14m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c5a89416ca894e62b4daf6f965981fde : SUCCESS in 2h 13m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bcfd856dfe2e4ccab6c1ed5c8740e56d : SUCCESS in 59m 42s","accounts_in_message":[],"_revision_number":1},{"id":"0f15ad14efe6a2d171de7c5af9ad3b67d003f804","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-22 18:46:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/64/865164/1/check/dsvm-tempest-cisco-zonemanager-job/0d27a81/ : RETRY_LIMIT in 8m 40s","accounts_in_message":[],"_revision_number":1},{"id":"902cd8c807e517844bf777313358bdec87459621","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-11-22 21:30:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/64/865164/1/ : SUCCESS in 1h 35m 07s","accounts_in_message":[],"_revision_number":1},{"id":"7b3930e7c44f45b12f7ec88e2e12fa53e4340c0c","author":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"date":"2022-11-24 06:15:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"388dd0a63053a93c6dab5d68ba59dd119f8ec550","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-24 06:28:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 7m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c011b2b3a182fd4fdfbf84a529f3984755cf2acd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 08:44:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/778674ea6c3d42a19a3095a9cf2252be\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d86de76a027c40358fc1c7cec8e831ed : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d3e1692b950416eabe7094e4c8f26d8 : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fc1a2540eea45a6b70fb8bc932fac80 : SUCCESS in 9m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f6a3b531c1f4ba99a8c7aea1f8bac00 : SUCCESS in 13m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/c6bdc3d18cd146678e6eb008df86d2c2 : SUCCESS in 1h 11m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a5ca178acb8345eaaf3d4aa29c4b3ca1 : SUCCESS in 1h 19m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57b948dddcdf402fa67c1ab61fb66478 : SUCCESS in 58m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26b623cf75614ec5b68971f0ebb0f255 : SUCCESS in 8m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c96591a34bf94945a0699b2578beaad4 : SUCCESS in 13m 10s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/661b41eb925042549e2baa4aa67bdbcf : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a4e777c0442143aba56c0fef966e1127 : SUCCESS in 4m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/021cd4e33afb489c87c8f5735f648e7b : SUCCESS in 5m 23s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5777b293a8e14bab9a08e8806c7559a1 : SUCCESS in 43m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9ab8b3410e1e46c8b61945bb45d254aa : FAILURE in 5m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/19bb70d9278e4fbcac499c7814e0ac37 : SUCCESS in 1h 22m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4101048879e045cabfddcb7d5e39b930 : FAILURE in 1h 53m 49s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0454436997db4388be111a55195d5f62 : SUCCESS in 1h 26m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bac810a6b7a746838c409b2bc8cc728f : SUCCESS in 1h 03m 04s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/47b291f6c6ca40ff94df171c775a619e : SUCCESS in 1h 14m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e6547a1d86ef48ac9f0737b8dff117d1 : SUCCESS in 1h 26m 39s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bffb8f58c7b847e68b636b1fcd6b8e8e : SUCCESS in 2h 23m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18922a4ad4454fd3a817f7994f0223b1 : SUCCESS in 43m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a655dce2e29bef1fbdd945bb86fa9d3a0e7fc030","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2022-11-28 07:38:45.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f3ad9baddd2fa2573c6769e1b9b5a620f66d2e2a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-12-12 06:43:19.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b68dd9f5a7d1dcfa86e396bec26604e38a19a38a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-12-12 18:41:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"66a16209cc6a77c6f1fb5d18a80671a33ec31509","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 18:42:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b38bf107b367c7d77ccc86754a645f82d4b1ebbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 22:06:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39395b8b7d714becbb2de479d644cb0a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15bde29b8a094388a2816e7cf87feefb : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3a4e0d71f9145fd8df00b6d506428d7 : SUCCESS in 10m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f908bf0d973746308e8f4dac5d6a26e2 : SUCCESS in 8m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be156f1412aa48f3ae40c0a9c4e0edc2 : SUCCESS in 15m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/0ddf18a4b1814557a76a2ba4422ed064 : SUCCESS in 54m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/27a6c36efec44db189e2924fa86accc5 : SUCCESS in 53m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9bcaa25261840399194783e0caaee9e : SUCCESS in 47m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7bbacffbb514b35937e0788dcbdb8fb : SUCCESS in 7m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/071bf2239b434d0286b3ea0b6d56de58 : SUCCESS in 59m 08s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f277f135c3634dd0a3931e18fd583a95 : SUCCESS in 1h 12m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8fa965bc2dc4dcf906fbe18c59a824a : SUCCESS in 51m 54s","accounts_in_message":[],"_revision_number":1},{"id":"43e92fbd5f3e75b16ec464e1637a5423a2e1d9bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 22:06:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"50905aafbd4d2034956ed20015ca50917c382692","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 22:07:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81fe9483a685488eaf44019f2e44827d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f25c0ff60c5a49e8ae400cc2a4c589e4 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e70b2155b9c42d0946777af9070157e : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d061455d4f7974a6f4c86ba7e54b4415be4a3385","revisions":{"d061455d4f7974a6f4c86ba7e54b4415be4a3385":{"kind":"REWORK","_number":1,"created":"2022-11-22 15:56:31.000000000","uploader":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"ref":"refs/changes/64/865164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/64/865164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/865164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/865164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/64/865164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/64/865164/1"}}},"commit":{"parents":[{"commit":"5cb0a8bb7d9c67ccc3bc44190170036fbc00a704","subject":"Merge \"Don\u0027t destroy existing backup by mistake on import\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5cb0a8bb7d9c67ccc3bc44190170036fbc00a704"}]}],"author":{"name":"Christophe Drevet-Droguet","email":"cdr@antemeta.fr","date":"2021-02-04 17:18:46.000000000","tz":60},"committer":{"name":"Manojkatari","email":"mkatari@redhat.com","date":"2022-11-22 15:56:31.000000000","tz":0},"subject":"[PURE] support IPv6 / add parameter pure_iscsi_cidr_list","message":"[PURE] support IPv6 / add parameter pure_iscsi_cidr_list\n\nPure storage: Add support for IPv6 in pure_iscsi_cidr parameter.\nPure storage: Add parameter pure_iscsi_cidr_list to support a list of networks.\n\nDepends-On: Ib001e2b4d1347754c2b46730bc10d86e8cdab7ad\nChange-Id: Ib04e49dac9fdabd81809ca389553ce3638bb9bbf\nCloses-Bug: 1910143\n(cherry picked from commit 7aa5e95d48c1bd0110001fb967c80e101f5b94dd)\n(cherry picked from commit 84dc4fb0599743ad19eac43ca2e10dedb5d5386b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d061455d4f7974a6f4c86ba7e54b4415be4a3385"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d061455d4f7974a6f4c86ba7e54b4415be4a3385"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
