)]}'
{"id":"openstack%2Fcinder~865416","triplet_id":"openstack%2Fcinder~master~Icf7a13dd942ef7d18838d3fd5db14b27c5214cf5","project":"openstack/cinder","branch":"master","topic":"865416","attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-11-23 11:32:53.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Icf7a13dd942ef7d18838d3fd5db14b27c5214cf5","subject":"Limit number of concurrent volume upload operations","status":"NEW","created":"2022-11-23 11:32:45.000000000","updated":"2022-11-30 03:45:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"597bb1588fc5dd56413e8c45cd7574a55de22b39","_number":865416,"virtual_id_number":865416,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-11-28 16:31:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"all":[{"value":1,"date":"2022-11-28 19:39:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-23 11:32:53.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-11-23 11:34:39.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-11-23 12:00:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-23 13:05:16.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-23 13:15:40.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-23 13:20:54.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-11-23 13:23:36.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-11-23 13:36:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-23 14:57:31.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-24 11:45:53.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-24 20:07:59.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2022-11-28 14:45:30.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-11-28 16:40:16.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-28 19:24:33.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2022-11-28 19:39:12.000000000","updated_by":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"reviewer":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"state":"REVIEWER"}],"messages":[{"id":"28104983da9dd30f01f3fe42a73fa0aef150d65b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-11-23 11:32:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d8d5146812b41b1714c7d8070e9754f8207844f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-23 11:32:53.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1771 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"4a63c88d0b78cfbe208d0f7d28f228985e0f4c6a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-23 11:34:39.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"4f183b7fb059ee71407b796baa1a635eaa9ce797","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 12:00:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5ac13942f5c4447b371cdf6351dd775\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/82a83d88c93e4e84ad90c54c6b4838b6 : FAILURE in 18m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/94ba6fdd7a3f44e9bf1f40b90b2e00b2 : FAILURE in 13m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49d24cc8c49307e20d988c4abeffc885b88aed1b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-23 13:05:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/865416 : SUCCESS in 1h 31m 01s","accounts_in_message":[],"_revision_number":1},{"id":"308189ad40209f0fe90085af38db1fdd359ea5ad","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-11-23 13:15:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/16/865416/1/ : SUCCESS in 1h 34m 30s","accounts_in_message":[],"_revision_number":1},{"id":"3922ec6625b3cc2ae398c11aa3178edeeafa97ac","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-23 13:20:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/16/865416/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/aed954e/ : SUCCESS in 1h 47m 00s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/16/865416/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/1c53bdc/ : SUCCESS in 1h 43m 55s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/16/865416/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/1e06ad3/ : SUCCESS in 1h 46m 01s","accounts_in_message":[],"_revision_number":1},{"id":"42cfb1a164857591164ce7133c9ef0a642e8fd4a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-23 13:23:36.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/16/865416/1/check/DellEMC_Unity_FC/b9581b8/DellEMC_Unity_FC/2026 : SUCCESS in 1h 11m 19s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/16/865416/1/check/DellEMC_Unity_iSCSI/e687ca6/DellEMC_Unity_iSCSI/1873 : SUCCESS in 1h 50m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"484934a3b100a5b18c3721450d6e484bbda5aa39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 13:36:50.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ae86b06a38844139f3668e77dc2c7ec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eee71d814c5445ef9a620a8be6e59b5a : FAILURE in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac938ab5ad4d461ca0f5e4b1a1c0e87b : FAILURE in 9m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81343caac9224c52b38a32670badf5ca : FAILURE in 10m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/768c7322948645a6ab8e28b4da8274a3 : SUCCESS in 14m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/7ed0fcfb983c4c3a8935cc8a7612e3e5 : SUCCESS in 49m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/279c8e5ea0ad48ada40332129f3fbe89 : SUCCESS in 1h 10m 12s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d227c3b2344f48f0b5a8088f6235cae5 : SUCCESS in 1h 15m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94fcf72431114f00947be5a9d0af6584 : SUCCESS in 49m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/54fd1f29f6834454953f523b1fbfe92a : FAILURE in 15m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3d158411a64e41b599adc9c64e0125d7 : SUCCESS in 5m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bbcd040178e04555889536fc9402c646 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ffd834d032014ef694d08d90fed8b3e6 : SUCCESS in 5m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d2a1b400579e431fbefe61002bf87261 : SUCCESS in 5m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f28429dea7ae41629049b67035c982e7 : SUCCESS in 38m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6aff596e68214c44a74060f8aa6bf1bc : FAILURE in 5m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6a0a8c92549f4a4ea8ea2c61b36e0dd4 : SUCCESS in 53m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/81aaa958558844cb84f84e0d9b35643e : FAILURE in 1h 10m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/480338fd1a2143b1bc1a4ccb845eea04 : SUCCESS in 1h 36m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f355ec3dab4640d2915d979f29f12472 : SUCCESS in 1h 10m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/602d05b5a28548feb91c11dabbca7f44 : SUCCESS in 1h 25m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9f4c7c2d441e4d0a843a1ecf3be14527 : SUCCESS in 40m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/adfee85d27464f5fb918979b2da32037 : SUCCESS in 1h 18m 05s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ecd627a95db742db8e460608884dc027 : SUCCESS in 2h 01m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68957e576b834c25acac5f362cc322f2 : SUCCESS in 36m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b6e174048d1741e264a7e9d273a5ac40c4a7599e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-23 13:42:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/865416 : FAILURE in 2h 07m 31s","accounts_in_message":[],"_revision_number":1},{"id":"50a206f361da4c1ea55343f58d8c82cd6dbc8fd7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-23 13:44:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/16/865416/1/check/iscsi-driver-alletra-6k/b4fd214/b4fd214.html : FAILURE in 2h 10m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c631723011611ecccfcd81389b448b7992019d74","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-23 14:57:31.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-16-865416-1/ns5_nfs : SUCCESS in 3h 24m 14s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-16-865416-1/ns5_iscsi : FAILURE in 11m 57s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"08747eb24f6ef1356f8fced46d48da2196db3ff8","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-11-24 11:45:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/16/865416/1/20221124171001/NEC-ISCSI : SUCCESS in 3h 35m 19s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/16/865416/1/20221124171001/NEC-FC : SUCCESS in 3h 35m 49s","accounts_in_message":[],"_revision_number":1},{"id":"221945491829853c667df9406b5ea42ff77c178d","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-11-24 20:07:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/16/865416/1/20221124225913/NEC-V-ISCSI : SUCCESS in 2h 37m 19s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/16/865416/1/20221124225913/NEC-V-FC : SUCCESS in 6h 08m 43s","accounts_in_message":[],"_revision_number":1},{"id":"73fa99453e929e65c65114aa3736c89d77ae0769","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-11-28 06:57:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"84bc3b5cfb2488d398625d75a05d63fe26378cc7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-28 06:57:15.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1799 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"86ab28c92a5116270e4c3d7ed2bec95d364af1fe","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-28 06:59:00.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"8334f2645854210dbe0338121618dcee9fdede1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-11-28 07:06:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2c26a5aed7ed8a1670a92e3a11e902cbc260242f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-28 07:06:15.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1800 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"aed38fa369b30968522ccd1cdc6eb145c282d78b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-28 07:11:04.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k finger://openstackcifc-virtual-machine/14bb8d07ba244a3282f0d95c6e586f75 : RETRY_LIMIT in 10s","accounts_in_message":[],"_revision_number":3},{"id":"702c363b5b3aa606b754033589532e43d044b68e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 07:29:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff78d4b4f3e9404fbc85a8436bb9bbac\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd25923f896e499bbd0bd8bd05530743 : FAILURE in 20m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/de0385370bbf4860ae31ac52770860ca : FAILURE in 20m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8f81acea16e074265a6aff7d7df9db52169e7927","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-28 08:38:03.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/865416 : SUCCESS in 1h 26m 33s","accounts_in_message":[],"_revision_number":3},{"id":"936596df1d1eda8a0e8c2630893ba57198400e20","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-28 08:39:22.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/865416 : SUCCESS in 1h 28m 14s","accounts_in_message":[],"_revision_number":3},{"id":"695da4bf45e61e24d72eac8c5b953fd67434df97","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-28 08:52:49.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/16/865416/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/96646e0/ : SUCCESS in 1h 32m 27s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/16/865416/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/3b28f47/ : SUCCESS in 1h 43m 51s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/16/865416/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/f6dae4e/ : SUCCESS in 1h 38m 29s","accounts_in_message":[],"_revision_number":3},{"id":"7dc3ef6aa60cfd37d37d7f13a8c5f205211db0f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 08:53:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0c645024b07466fb3e5d2667a7e6cc6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59458708d61949ef8e1f0a8b28eea158 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43afbb9fcf304078bb4416cee5a0d2c9 : FAILURE in 9m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f81d095dfde472c867b4f9534bbb2e7 : FAILURE in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/700891f5f59d49ffaa6138c5c7127fa3 : SUCCESS in 13m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/4249060a9e174821b98114fd56eaab9d : SUCCESS in 51m 54s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/771eb22145bc4b1f8ed2821c57d51a15 : SUCCESS in 53m 48s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fdc9b882a1104b87a8522944e0fc24bd : SUCCESS in 1h 22m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06137ca8ec654931a3ef08715735cc04 : SUCCESS in 49m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3d89cc622744441f97c272084ac7c1ac : FAILURE in 10m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f9b252d8b1674e94a39558507d161dc1 : SUCCESS in 5m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/efb0f363411248728a72228b58dd6f6f : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5de28a779ca647418594711069fe303e : SUCCESS in 6m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e200760cc11b4bcca86ad8036acde1d2 : SUCCESS in 5m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f426885ae07a4bdb97ce83e5a1212e4a : SUCCESS in 39m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b8f479fce4ff481d8ae5fcd7e6732ce7 : FAILURE in 5m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1f5df93fa13a41d9be76978ff2e42dea : SUCCESS in 1h 07m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/44066b668d284c69a4854f0c31882287 : FAILURE in 1h 22m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/05436ca3bb344dbdb410a324a143d400 : SUCCESS in 1h 42m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/23cc949a3c204a629ea1812e5fdb4e35 : SUCCESS in 46m 11s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ba0a4000fcf64102b97048eefbeec145 : SUCCESS in 1h 29m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4a9ea059fcd64bc9a25c0f79af4457e9 : SUCCESS in 31m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4b909444c153408db5096c12e044bcf8 : SUCCESS in 1h 24m 39s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b1cb944980114982beaf5b92ef2fee75 : SUCCESS in 1h 41m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/386a5533d55c49bdb4ace1be3346760f : SUCCESS in 45m 45s","accounts_in_message":[],"_revision_number":3},{"id":"62de6aa630faed14d0bbc3944fafe1c5f9fa8d07","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-28 09:07:50.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/16/865416/3/check/iscsi-driver-nimble/1330425/1330425.html : FAILURE in 1h 05m 31s","accounts_in_message":[],"_revision_number":3},{"id":"5faf484e57fa1b1b9c13695b8abc75fbfe5ccddf","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-11-28 10:34:46.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/16/865416/3/ : SUCCESS in 1h 36m 38s","accounts_in_message":[],"_revision_number":3},{"id":"862d757b323fbc80009062409954b6aafd839a77","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-28 10:51:04.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-16-865416-2/ns5_nfs : SUCCESS in 3h 29m 47s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-16-865416-2/ns5_iscsi : SUCCESS in 3h 31m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"3c821ba00a1b92471f6b931a45efef594a1f765c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-28 11:13:51.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-16-865416-3/ns5_nfs : SUCCESS in 3h 30m 20s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-16-865416-3/ns5_iscsi : FAILURE in 3h 31m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"165de55d5cbce657607412044d2940aeb30e965d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-11-28 11:42:10.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/16/865416/3/20221128171050/NEC-ISCSI : SUCCESS in 3h 30m 32s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/16/865416/3/20221128171050/NEC-FC : SUCCESS in 3h 31m 18s","accounts_in_message":[],"_revision_number":3},{"id":"9ab3c6b972787478869f80817fc985904d633638","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-11-28 14:22:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f1796666b6e144b7ba30f286c344df302ea37633","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-28 14:22:45.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1802 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"bbfdfc8e58eab925ddd8192d75c42e6f20b5ec88","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-28 14:45:30.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 7m 30s","accounts_in_message":[],"_revision_number":4},{"id":"402b936b86c85c8032612b0d8f9de3efafbbfeef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 14:50:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/312ff99b3b174452a6c9ea2b2ffd58dd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ef58654dc8bc4cf0b3d779ffba5ec60e : SUCCESS in 23m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2a1f448f31f14bf2af7ca128d0a7e566 : SUCCESS in 25m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"93824d3e1aea456a8c125b13ab5321edb973b0c4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-28 15:40:32.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/16/865416/4/check/iscsi-driver-nimble/493f4e2/493f4e2.html : FAILURE in 1h 07m 01s","accounts_in_message":[],"_revision_number":4},{"id":"da6ffb81ac886d87c3a93ab77ecee2a7533310ce","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-28 15:58:45.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/865416 : SUCCESS in 1h 33m 04s","accounts_in_message":[],"_revision_number":4},{"id":"3d650be897b17ee88f1d29e816fd80490b8f4ce5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-28 16:00:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/865416 : SUCCESS in 1h 35m 51s","accounts_in_message":[],"_revision_number":4},{"id":"93d3bd5ca67dad2dce022d738127c7c980993ea3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-28 16:09:56.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/16/865416/4/check/iscsi-driver-alletra-6k/539b82b/539b82b.html : FAILURE in 1h 44m 28s","accounts_in_message":[],"_revision_number":4},{"id":"145389c6013c3a119e3d4163bba22bf09d3a1d6a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-28 16:24:19.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/16/865416/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/0dccb3a/ : SUCCESS in 1h 45m 18s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/16/865416/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/bd2e2c1/ : SUCCESS in 1h 49m 34s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/16/865416/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/9d7604b/ : SUCCESS in 2h 00m 24s","accounts_in_message":[],"_revision_number":4},{"id":"99658602067778edd84fc3ae86d73bfc6e4147bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 16:31:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0323507efd4140e3bfeeab0f1946641e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d872830f284428c98adbad3c6344f32 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d48ae98f2026458398fdc2a178116109 : SUCCESS in 12m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a9570cf960a45fca9e3eb83315c0f4d : SUCCESS in 10m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df0119eb50d64e2684635b4b78487c9c : SUCCESS in 12m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/0a144fccc3d5414fa04893abe0d3caf2 : SUCCESS in 54m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4988026077c54893a992f850e2270361 : SUCCESS in 1h 02m 40s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/beefff4c70f749ca9c7fa2d63557fb8b : SUCCESS in 1h 21m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63c8750bfdf1468a94259d5f56deba08 : SUCCESS in 42m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcee91d38de6490a89f9e2ae4b2c27a2 : SUCCESS in 8m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3209d2209f6641f0a193fd5ab9b52309 : SUCCESS in 14m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0353211bfe554459ab141a7dc3213949 : SUCCESS in 4m 39s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3d335ffbf27c4c24845d78544a2225f1 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5a50f3a3e53b42cd8b8f4e4c76019283 : SUCCESS in 4m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a41ba6dc82234480acfe9147dd88824b : SUCCESS in 5m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b34bc18297294f178bf544f4efa8f2e8 : SUCCESS in 51m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cee44c3d34594a7aa9360cd14338fa89 : FAILURE in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b0768439a88d4f6d82f181a9dc57fbaa : SUCCESS in 1h 00m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/319fc534c7b64f6089fadba2b44f2a29 : FAILURE in 1h 07m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e6ce9f7a0d7140b2bd315b4edc121a55 : SUCCESS in 1h 42m 43s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f5f42f3258ad40819446b3e6f7de6959 : SUCCESS in 52m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d8e24d693c334463872631d653ab2d56 : SUCCESS in 1h 13m 12s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/904a54169fde4fb0a5173640f12e0640 : SUCCESS in 33m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4881961561924a109e893a6ea2e52cbe : SUCCESS in 1h 15m 25s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a5ffee618062465baed721ae761445d5 : SUCCESS in 1h 55m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6f44e1846bc42b4881575aef370558c : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":4},{"id":"59d4e1ab4bd775c8eee89a9c1a3705d6d9209fdb","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-11-28 16:40:16.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/16/865416/4/check/fujitsu-eternus-dx-iscsi/1a64faa : FAILURE in 2h 14m 41s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/16/865416/4/check/fujitsu-eternus-dx-fc/49138ec : FAILURE in 2h 17m 01s","accounts_in_message":[],"_revision_number":4},{"id":"2562319e50801eae90f0765eafdca4c82ace5400","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-11-28 17:42:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/16/865416/4/ : SUCCESS in 1h 46m 46s","accounts_in_message":[],"_revision_number":4},{"id":"459cd4bcfdd7458a1eebc790c9b95dcb19bc77cf","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-28 17:54:46.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-16-865416-4/ns5_nfs : SUCCESS in 3h 27m 9s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-16-865416-4/ns5_iscsi : SUCCESS in 3h 31m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"b7cb874cfa0ce5c21dc3d66d36a3a865a6bd0af3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-11-28 18:50:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/16/865416/4/20221129001640/NEC-ISCSI : SUCCESS in 3h 33m 02s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/16/865416/4/20221129001640/NEC-FC : SUCCESS in 3h 33m 23s","accounts_in_message":[],"_revision_number":4},{"id":"743269a9f9c66d490d7a43c61d03b5809988f155","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2022-11-28 19:24:33.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n* Infinidat-FC http://openstack-ci-logs.aws.infinidat.com/#logs/refs-changes-16-865416-4/FC/ : SUCCESS in 1h 58m 7s\n* Infinidat-iSCSI http://openstack-ci-logs.aws.infinidat.com/#logs/refs-changes-16-865416-4/iSCSI/ : SUCCESS in 1h 31m 59s\nFor more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"9365552cbedd8f891bc072870a4339cc40dd13e6","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2022-11-28 19:39:12.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f25395175afbaf0e86081a9f3a7e7cff1a2a9012","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-11-29 22:01:05.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/16/865416/4/20221130002726/NEC-V-FC : SUCCESS in 5h 38m 52s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/16/865416/4/20221130002726/NEC-V-ISCSI : SUCCESS in 6h 33m 36s","accounts_in_message":[],"_revision_number":4},{"id":"597bb1588fc5dd56413e8c45cd7574a55de22b39","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-30 03:45:11.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/16/865416/4/check/DellEMC_Unity_FC/24ad577/DellEMC_Unity_FC/2041 : SUCCESS in 1h 15m 04s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/16/865416/4/check/DellEMC_Unity_iSCSI/065ab14/DellEMC_Unity_iSCSI/1888 : FAILURE in 1h 58m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ccbb355ff92595b36b3536a6a755ed3760515e07","revisions":{"8dca71f867c1f7509219d7acfe6e4e0f1d84cf1c":{"kind":"REWORK","_number":1,"created":"2022-11-23 11:32:45.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/16/865416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/865416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/865416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/865416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/865416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/865416/1"}}},"commit":{"parents":[{"commit":"7c1a5ce7b11964da4537fd6a7d157ede646b9e94","subject":"Merge \"nit: correct storwize_portset help text\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7c1a5ce7b11964da4537fd6a7d157ede646b9e94"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-23 11:25:10.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-23 11:32:12.000000000","tz":330},"subject":"WIP: Limit number of concurrent volume upload operations","message":"WIP: Limit number of concurrent volume upload operations\n\nIn cases where backend takes huge amount time and IO bandwidth,\nit could add load to the system and hence cause issues.\nLike in case of rbd, we export the volume first which, depending\non the number of concurrent operations, can take upto several\nminutes and a lot of IO/bandwidth. this also increases load on\nthe image conversion dir if there are other operations also running\non that directory.\n\nChange-Id: Icf7a13dd942ef7d18838d3fd5db14b27c5214cf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8dca71f867c1f7509219d7acfe6e4e0f1d84cf1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8dca71f867c1f7509219d7acfe6e4e0f1d84cf1c"}]},"branch":"refs/heads/master"},"da5b1d5f161fa2c9ff51e0a5b3cb1793eeabff70":{"kind":"REWORK","_number":2,"created":"2022-11-28 06:57:08.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/16/865416/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/865416/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/865416/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/865416/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/865416/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/865416/2"}}},"commit":{"parents":[{"commit":"7c1a5ce7b11964da4537fd6a7d157ede646b9e94","subject":"Merge \"nit: correct storwize_portset help text\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7c1a5ce7b11964da4537fd6a7d157ede646b9e94"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-23 11:25:10.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-11-28 06:57:29.000000000","tz":0},"subject":"WIP: Limit number of concurrent volume upload operations","message":"WIP: Limit number of concurrent volume upload operations\n\nIn cases where backend takes huge amount time and IO bandwidth,\nit could add load to the system and hence cause issues.\nLike in case of rbd, we export the volume first which, depending\non the number of concurrent operations, can take upto several\nminutes and a lot of IO/bandwidth. this also increases load on\nthe image conversion dir if there are other operations also running\non that directory.\n\nChange-Id: Icf7a13dd942ef7d18838d3fd5db14b27c5214cf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da5b1d5f161fa2c9ff51e0a5b3cb1793eeabff70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/da5b1d5f161fa2c9ff51e0a5b3cb1793eeabff70"}]},"branch":"refs/heads/master"},"0f1cd5f8c41388d9321cf6ff857e28b8d968bdbd":{"kind":"REWORK","_number":3,"created":"2022-11-28 07:06:05.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/16/865416/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/865416/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/865416/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/865416/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/865416/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/865416/3"}}},"commit":{"parents":[{"commit":"7c1a5ce7b11964da4537fd6a7d157ede646b9e94","subject":"Merge \"nit: correct storwize_portset help text\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7c1a5ce7b11964da4537fd6a7d157ede646b9e94"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-23 11:25:10.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-11-28 07:06:12.000000000","tz":0},"subject":"WIP: Limit number of concurrent volume upload operations","message":"WIP: Limit number of concurrent volume upload operations\n\nIn cases where backend takes huge amount time and IO bandwidth,\nit could add load to the system and hence cause issues.\nLike in case of rbd, we export the volume first which, depending\non the number of concurrent operations, can take upto several\nminutes and a lot of IO/bandwidth. this also increases load on\nthe image conversion dir if there are other operations also running\non that directory.\n\nChange-Id: Icf7a13dd942ef7d18838d3fd5db14b27c5214cf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0f1cd5f8c41388d9321cf6ff857e28b8d968bdbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0f1cd5f8c41388d9321cf6ff857e28b8d968bdbd"}]},"branch":"refs/heads/master"},"ccbb355ff92595b36b3536a6a755ed3760515e07":{"kind":"REWORK","_number":4,"created":"2022-11-28 14:22:38.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/16/865416/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/16/865416/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/865416/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/865416/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/16/865416/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/16/865416/4"}}},"commit":{"parents":[{"commit":"7c1a5ce7b11964da4537fd6a7d157ede646b9e94","subject":"Merge \"nit: correct storwize_portset help text\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7c1a5ce7b11964da4537fd6a7d157ede646b9e94"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-23 11:25:10.000000000","tz":330},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2022-11-28 14:20:42.000000000","tz":330},"subject":"Limit number of concurrent volume upload operations","message":"Limit number of concurrent volume upload operations\n\nIn cases where backend takes huge amount time and IO bandwidth,\nit could add load to the system and hence cause issues.\nLike in case of rbd, we export the volume first which, depending\non the number of concurrent operations, can take upto several\nminutes and a lot of IO bandwidth. After that we use python IO to\ncopy that exported volume to glance which again is IO consuming\noperation. This also increases load on the image conversion dir\nif there are other operations also running on that directory.\n\nChange-Id: Icf7a13dd942ef7d18838d3fd5db14b27c5214cf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ccbb355ff92595b36b3536a6a755ed3760515e07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ccbb355ff92595b36b3536a6a755ed3760515e07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
