)]}'
{"id":"openstack%2Fcinder~866093","triplet_id":"openstack%2Fcinder~master~Ie58db6f70aac0b13fbe832ecee177146e5dda397","project":"openstack/cinder","branch":"master","topic":"ceph_backup_mypy","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2024-07-26 19:37:50.000000000","reason":"Change was marked work in progress"},"34149":{"account":{"_account_id":34149,"name":"Niklas Schwarz","email":"niklas.schwarz@inovex.de","username":"nschwarz"},"last_update":"2024-07-26 19:37:50.000000000","reason":"Change was marked work in progress"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-07-26 19:37:50.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Ie58db6f70aac0b13fbe832ecee177146e5dda397","subject":"DNM mypy, Ceph: correct types of file-likes in backup and restore","status":"NEW","created":"2022-11-29 23:25:49.000000000","updated":"2024-07-29 01:38:28.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":6,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"dede992a0d256aa6171c249c49e37c8cf866af5d","_number":866093,"virtual_id_number":866093,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-07-26 21:44:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34149,"name":"Niklas Schwarz","email":"niklas.schwarz@inovex.de","username":"nschwarz"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}],"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-11-29 23:25:59.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-11-29 23:30:53.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-11-29 23:50:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-29 23:50:15.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2022-11-29 23:59:44.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-30 00:12:52.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2022-11-30 01:17:16.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-30 01:30:08.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-30 01:34:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-30 01:38:23.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-11-30 02:58:06.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-30 03:46:06.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-30 07:40:36.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-12-01 12:28:07.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-04-05 06:58:55.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":34149,"name":"Niklas Schwarz","email":"niklas.schwarz@inovex.de","username":"nschwarz"},"state":"CC"},{"updated":"2024-04-05 06:58:55.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2024-07-26 19:37:43.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2024-07-26 20:59:34.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-07-26 22:33:38.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-26 22:39:31.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2024-07-26 23:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"}],"messages":[{"id":"94b696dec6180855f85d3a3191f5e05346cf2ed2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-11-29 23:25:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a62461dbfe7e5ebc2a9a3f080f74090d76d7d772","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-11-29 23:25:59.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1809 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"409a9617a19427729640ffc13a71ca91838a32d5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-29 23:30:53.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"04480f70025ff75af2f3c77b8abd2f064f671297","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 23:50:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ead8a174a9c44c087b09ff28e2c1f48\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2664c7e0802640d2af9a0dd83d892dbc : SUCCESS in 21m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1f185cacafbf4ed9ba1b2278ccfb6f81 : SUCCESS in 21m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"87124ff86bcf276a9da50e63eb7a0045d15c79b0","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-11-29 23:54:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d3f6299fe6901d559a654575f330568098c475db","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-29 23:59:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/866093 : FAILURE in 32m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e95697053eade72fde9bd81fbdabdf04b1138611","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2022-11-30 00:12:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : ERROR Failed to update project None in 7m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d456dd2052b10c235917c47552c9acb86f1aaa88","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-11-30 01:03:26.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/866093 : FAILURE in 1h 35m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f8cc739e3fcc8e9dedb5c1ee194d405eee686e45","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-11-30 01:17:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/93/866093/1/check/fujitsu-eternus-dx-iscsi/9fdce13 : FAILURE in 1h 46m 54s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/93/866093/1/check/fujitsu-eternus-dx-fc/8082708 : FAILURE in 1h 50m 49s","accounts_in_message":[],"_revision_number":1},{"id":"feb849b763f59ed4c5e1a638ea358ce0d8be44a8","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-11-30 01:30:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/93/866093/1/ : SUCCESS in 1h 55m 57s","accounts_in_message":[],"_revision_number":1},{"id":"fd4097e724e201900b4061d4eb8e42a9e8ede38b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 01:34:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb3069c4b3df4585b13ca6b48b309c74\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba8058680eeb4352b8a465366fbcc891 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/325d4e61882c446394d44b10bd97a5dc : SUCCESS in 11m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2bd36ec6b704f62978306aeb07348a2 : SUCCESS in 12m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbe2de22666c4e00be42b28225b20c13 : SUCCESS in 13m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/1ac2b88a906f46789b0a033976d760a9 : SUCCESS in 59m 12s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f643e8cbf23444f49bf1303a4d40ec61 : SUCCESS in 40m 54s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e876473f0c26408ab2d1fd331b90abbe : SUCCESS in 1h 30m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e15112f946b84e979e9d5a475d84706e : SUCCESS in 1h 02m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/52490ebea5c04e44926ef4f7ccd247ae : SUCCESS in 16m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/561ff9475d194937993dccd6d6c1f6ba : FAILURE in 5m 17s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f7bb482863b4453bbd19d2456a8c951f : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3e80babd892547e2b8864cf9bb6dd9ae : SUCCESS in 5m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a3df9ff8805d406e9e6f0a572d5f9d70 : SUCCESS in 5m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/548869baaa5f43dca28dd6489df06b6b : SUCCESS in 51m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0796d9cfb67a4e938c75fdb157da9f56 : FAILURE in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/20d8c9df83d148c1bbc0aa7d190f65f7 : SUCCESS in 1h 16m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8d5191d286f543c5b4065af9e9a9e08e : FAILURE in 1h 21m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/db8c88e2c3464c3a89953d20e1484907 : SUCCESS in 2h 01m 48s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3907d9b8457e4b36ba0e9c0a43cf6d60 : SUCCESS in 1h 01m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/f9c3b7b7cea347ac88a8e57bfc4288ac : SUCCESS in 1h 36m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/493e8086054044a3a9df02d82eb89a03 : SUCCESS in 43m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/031b0c4794dc482880d3ef57a1db5b23 : SUCCESS in 1h 22m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bbab13ec992b4899806f52ce41ae0b48 : SUCCESS in 1h 38m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d458dbf0b774bda96e289e06c7455d9 : SUCCESS in 50m 01s","accounts_in_message":[],"_revision_number":1},{"id":"58b9f2113b16ffd9bd0b76009d8c718312fbe02d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-30 01:38:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/93/866093/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/c2a388d/ : SUCCESS in 1h 35m 19s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/93/866093/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/2e4d52c/ : SUCCESS in 2h 01m 02s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/93/866093/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/e25bd8a/ : FAILURE in 2h 11m 30s","accounts_in_message":[],"_revision_number":1},{"id":"3f33480753e82038d41c7678a249909327250c51","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-11-30 01:40:43.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/93/866093/1/check/iscsi-driver-alletra-6k/96b0862/96b0862.html : FAILURE in 2h 12m 49s","accounts_in_message":[],"_revision_number":1},{"id":"1efaa5b6319bf6754f4efaf50ef8ee5c5b2fcb21","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-30 02:58:06.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-93-866093-1/ns5_nfs : SUCCESS in 3h 28m 47s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-93-866093-1/ns5_iscsi : SUCCESS in 3h 31m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"6bfd92e9f307aaa51639f9c8dbecaba0a9ef2d20","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-11-30 03:46:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/93/866093/1/20221130091338/NEC-ISCSI : SUCCESS in 3h 31m 31s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/93/866093/1/20221130091338/NEC-FC : SUCCESS in 3h 32m 24s","accounts_in_message":[],"_revision_number":1},{"id":"efa9bb595a33ef4c94beb8e1ef3bb8f4a732f209","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-30 07:40:36.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/93/866093/1/check/DellEMC_Unity_FC/349988b/DellEMC_Unity_FC/2043 : SUCCESS in 1h 22m 40s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/93/866093/1/check/DellEMC_Unity_iSCSI/4fef30f/DellEMC_Unity_iSCSI/1890 : SUCCESS in 1h 51m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"426bb40c5062b414e1520a5e8a90550a711e064b","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-12-01 12:28:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/93/866093/1/20221201145210/NEC-V-FC : SUCCESS in 5h 44m 27s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/93/866093/1/20221201145210/NEC-V-ISCSI : SUCCESS in 6h 35m 54s","accounts_in_message":[],"_revision_number":1},{"id":"846737bdf13d0fd583d3ba603e855bc58f69d127","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-12-02 00:55:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"39847b3a6646e8704259bec615a12423e929b869","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-12-02 00:56:04.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1825 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"99338303ed8e5d43dce359b924411d84101e5e18","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-02 01:14:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : MERGER_FAILURE in 4m 48s","accounts_in_message":[],"_revision_number":2},{"id":"117094479ed847fba89a9d60db146526c9ebd733","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 01:23:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fcbf6de47314fc0b638782a484934d4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f68eada35aa43a486a60050e76544f6 : FAILURE in 12m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/521559a865294afda304441d82b73606 : SUCCESS in 24m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2d04ffd64581dc8bea4c407c20e716f22e8e1670","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-02 02:06:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/93/866093/2/check/iscsi-driver-alletra-6k/b69516a/b69516a.html : FAILURE in 1h 09m 09s","accounts_in_message":[],"_revision_number":2},{"id":"41ed32d1c64ae17363c2b7d425a214cee2c2c78e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-02 02:25:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/866093 : SUCCESS in 1h 26m 15s","accounts_in_message":[],"_revision_number":2},{"id":"8a76f232aa5b596ded648ab5a95be8b3bc452b65","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-02 02:36:05.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/866093 : FAILURE in 1h 36m 11s","accounts_in_message":[],"_revision_number":2},{"id":"a8b51ca65b1bb403b0cb778e7f855f52124bddfb","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2022-12-02 02:54:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/93/866093/2/ : SUCCESS in 1h 49m 39s","accounts_in_message":[],"_revision_number":2},{"id":"61f627c079e6a69f7cb9a2218ae6d977a3a9ed06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 03:28:44.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/660ea6f371d24fd096f7da3347e3cb20\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60395601bf504c5b9d35eecd2441920b : FAILURE in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5e0c39503f247d4874e80842b25023a : FAILURE in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14a32195988a4af082912b96fd2c40a4 : SUCCESS in 10m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b46420e83e1845c6af1a6ec87e3d686b : SUCCESS in 13m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/8e3ffed6eb3745648a5a9f58da3f180a : SUCCESS in 1h 05m 36s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fd67612f502e4e39a4d99a75a6e25814 : SUCCESS in 51m 56s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d4d3774e9f524a3b82a4461a63f65f31 : SUCCESS in 1h 35m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07741ff30b2b4751b7da01fc7ad48b86 : SUCCESS in 48m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5210c6c4c63d47079c5c05e51ee80d1f : SUCCESS in 16m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/713a9be5f9044a75a68d103149a57024 : SUCCESS in 5m 58s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/088fb111147d4c2b9c9191fbd7f46b5d : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3cba24bff69649d9a5a5d0164e93336e : SUCCESS in 6m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6ddd676e8bc64821869603c2af855809 : SUCCESS in 6m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/22537c9f0a1b4b5186f0f0b4e18cca0f : SUCCESS in 46m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bc591f06fc57411f98fc566046ac281e : FAILURE in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/396f172c1ff34c43bb70a108e23c190e : FAILURE in 23m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3abf370b430d4309823f10293cd9822a : FAILURE in 28m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/78bba49cde9249f3909056620f2ab8dc : SUCCESS in 1h 48m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5724711854a84d00998dd1ec226dd922 : SUCCESS in 1h 02m 34s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d5d3f502abcb48cfa8fcf240e1e2e300 : SUCCESS in 1h 23m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/28fae98e8f7c4b89b96a59f52d6867b9 : SUCCESS in 33m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b4b28b57306a472b8926e8a49f748b0f : SUCCESS in 1h 29m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d4bf4fc1a97c4dd58367e6d798af549b : SUCCESS in 2h 25m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/633175fc270d4a8b8221b3c85bbfc501 : SUCCESS in 42m 50s","accounts_in_message":[],"_revision_number":2},{"id":"328a31f843f34428611682aba49dcdddcb5bfde8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-12-02 04:27:50.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-93-866093-2/ns5_nfs : SUCCESS in 3h 31m 23s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-93-866093-2/ns5_iscsi : SUCCESS in 3h 29m 10s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"3f1e8a620d325a4462ad78cccb50a9aa4ddf1248","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-02 04:31:34.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/93/866093/2/check/DellEMC_Unity_FC/b44ac15/DellEMC_Unity_FC/2058 : SUCCESS in 1h 50m 04s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/93/866093/2/check/DellEMC_Unity_iSCSI/97ce208/DellEMC_Unity_iSCSI/1905 : SUCCESS in 1h 56m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"f985070c21ad67f98a0af1d45231913a09bc13d5","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-12-02 15:59:27.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"16746626128d8f498be26d1a73cc29c27ee2177d","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-12-02 19:01:31.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"dc8ddc63992b5d0127946657180faa3384eab8bb","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-12-03 09:02:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/93/866093/2/20221203143019/NEC-ISCSI : SUCCESS in 3h 31m 40s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/93/866093/2/20221203143019/NEC-FC : SUCCESS in 3h 32m 35s","accounts_in_message":[],"_revision_number":2},{"id":"ae24e983ab1df8d1f179484ceadf906f84d30136","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-12-04 18:36:20.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/93/866093/2/20221204211303/NEC-V-ISCSI : SUCCESS in 3h 38m 01s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/93/866093/2/20221204211303/NEC-V-FC : SUCCESS in 6h 23m 14s","accounts_in_message":[],"_revision_number":2},{"id":"d5e61b638948d5fe071097c2760f4b5655c9406f","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2024-04-05 07:09:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"583610a40dc2fb32b6975ee5103113779f6b4deb","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-04-09 16:17:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"acee04197d7f41dedda5ca5b7891ade81701c7c9","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-07-26 19:36:44.000000000","message":"Patch Set 2: Workflow-1\n\n(2 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"31da0442cc81bc9d8a0c8c418bc3d5729a905a14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-07-26 19:37:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"fff7fb5de2f2ae9509de644c232ccc9b02312f93","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-26 19:37:43.000000000","message":"Patch Set 3:\n\nBuild Started http://9.180.170.18:8080/job/CI_Run/168/","accounts_in_message":[],"_revision_number":3},{"id":"5f7a4d47d9cc8f5796ad4415019997a57ebd9c2a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2024-07-26 19:37:44.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n\\* CI_Run http://9.180.170.18:8080/job/CI_Run/168/ : FAILURE (skipped)","accounts_in_message":[],"_revision_number":3},{"id":"b563903778643b2b09028567dde979789af8de86","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-07-26 19:37:50.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"5473ead6a048630e76169e3b558b663f39bae92e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 19:48:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22757a7203014d8a8c95d06c7736b6d5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/29ddc3cbc1644894923e13a41aa218b3 : SUCCESS in 9m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c64cdf07bd2644d18d8e16ea827314b1 : SUCCESS in 9m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c08cea0c804cc197027d686fa08716e595b1e0de","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-07-26 20:00:02.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/9522dc23b61f4b44a4d742bf9506d325\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/30f4502100ba4f1696d6964be0793097 : POST_FAILURE in 19m 57s","accounts_in_message":[],"_revision_number":3},{"id":"a856612940526d72a6934cdc545070da6202e634","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-07-26 20:28:57.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/866093 : SUCCESS in 42m 23s","accounts_in_message":[],"_revision_number":3},{"id":"f27d109569ff5c7ef520b615bc6d0bf5177057d9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-07-26 20:39:32.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/866093 : SUCCESS in 1h 00m 44s","accounts_in_message":[],"_revision_number":3},{"id":"269bc0b22c0c11e166cdb1e5f90e99331fcab4b1","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-07-26 20:59:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/806/.","accounts_in_message":[],"_revision_number":3},{"id":"cfbde8d8527f4670b7b9f1973296e933f25a2210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 21:44:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8aff646f398042658348563bf3842ecf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfad919a6cfb4302ac99831f8735b10b : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03346f59a858475ca8e1b044dc04ac4f : SUCCESS in 9m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45275971a2ef4d7ab1a0bdc665d9ce5c : SUCCESS in 8m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c0058a458da44928fd2929129794d72 : SUCCESS in 12m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a7e05a44927471e9d0eca0a8b905958 : SUCCESS in 12m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/60f5a0c704e745af9d610d8c3acde7d7 : SUCCESS in 1h 14m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7760aa9df6334319b242ea34841eecc1 : SUCCESS in 1h 24m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db1de416fd3a4791848a4ed64a298e71 : SUCCESS in 46m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d9196b80916d4f5d909b6514ec044d82 : SUCCESS in 17m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8e0a4687cd15477d89afd9be626323e7 : FAILURE in 4m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/19092e86133545c286d2ed8ead9a70af : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/87e7516ad7af43bc902dc987f432f702 : SUCCESS in 6m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7f1d6de884d6499390fa88cefc969210 : SUCCESS in 5m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5146abaed87f42f4a90bfe297f0c6eb6 : FAILURE in 24m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/864c868e476d4bb98e4983a5cff268c5 : SUCCESS in 6m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6b1f78e876884c0c8984a8382d177066 : SUCCESS in 1h 42m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dcf3900ac74942c2a575ce094b0e20c4 : SUCCESS in 2h 03m 45s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/adab1d9cb38a41718c78a91305c9b2ef : SUCCESS in 1h 37m 36s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/91011d3dacd546c7b86ca015f313be35 : FAILURE in 21m 26s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/3a0c3e44776445f9b407105d968b4c70 : SUCCESS in 36m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9430d5295772492886acc48a455023be : SUCCESS in 1h 15m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/864851ac536f417d8beec76b0ed95ab7 : SUCCESS in 1h 29m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5878cdc99ffa4e6cb2b69667433baec7 : SUCCESS in 36m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/06aac6b2a46b465bbf15c6e92aafaad8 : SUCCESS in 1h 25m 39s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/dbba936938ee47b1b0c2ce5ecc79ecab : FAILURE in 10m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a5f704e000cf41ea829a5f54e779f3c3 : SUCCESS in 1h 50m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b18d298c76846e683c838e4e19826c1 : SUCCESS in 46m 16s","accounts_in_message":[],"_revision_number":3},{"id":"8f2ef66ac87fec18b5c250d51ac09abc2cb5f635","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-26 21:49:44.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/93/866093/3/check/DellEMC_Unity_FC/b57301e/DellEMC_Unity_FC/2311 : SUCCESS in 1h 46m 06s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/93/866093/3/check/DellEMC_Unity_iSCSI/174639f/DellEMC_Unity_iSCSI/2316 : SUCCESS in 2h 11m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"27083722305297570d1ad05a8c50dbeaddb8a6db","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-26 22:33:38.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/93/866093/3/check/DellEMC_PowerFlex_3.6/2626adf/DellEMC_PowerFlex_3.6/1797 : SUCCESS in 1h 54m 04s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/93/866093/3/check/DellEMC_PowerFlex_4.5/64f4e69/DellEMC_PowerFlex_4.5/676 : FAILURE in 2h 02m 56s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/93/866093/3/check/DellEMC_PowerFlex_4.x_NVMeTCP/fae9f2d/DellEMC_PowerFlex_4.x_NVMeTCP/640 : FAILURE in 13m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"de2c2618d65fc21efbad96c3083c3c0e6fe685f4","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2024-07-26 22:39:31.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-fc-driver https://github.com/hpe-storage/hpe-ci-log/tree/master/logs/866093 : TIMED_OUT in 3h 00m 36s","accounts_in_message":[],"_revision_number":3},{"id":"79e5dede87aed65895b6bbb810880f0a4e7f4045","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-07-26 23:00:06.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/93/866093/3/fc : SUCCESS in 0h 43m 36s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/93/866093/3/iscsi : SUCCESS in 0h 58m 52s","accounts_in_message":[],"_revision_number":3},{"id":"9f58aee7d1150763f7c7b0d08040d9ea69a63614","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-07-27 16:39:17.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/93/866093/3/20240727212141/NEC-V-ISCSI : SUCCESS in 4h 04m 55s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/93/866093/3/20240727212141/NEC-V-FC : SUCCESS in 4h 17m 32s","accounts_in_message":[],"_revision_number":3},{"id":"dede992a0d256aa6171c249c49e37c8cf866af5d","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-07-29 01:38:28.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/93/866093/3/20240729021246/NEC-ISCSI : SUCCESS in 5h 15m 52s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/93/866093/3/20240729021246/NEC-FC : SUCCESS in 8h 25m 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"68ef026ebe5dad25568a3f3dca804f880e2d40f4","revisions":{"6eedb9d54b71ee0bfdda56e44c3f95571185c7ba":{"kind":"REWORK","_number":1,"created":"2022-11-29 23:25:49.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/93/866093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/866093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/866093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/866093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/866093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/866093/1"}}},"commit":{"parents":[{"commit":"7c1a5ce7b11964da4537fd6a7d157ede646b9e94","subject":"Merge \"nit: correct storwize_portset help text\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7c1a5ce7b11964da4537fd6a7d157ede646b9e94"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2022-11-29 23:18:54.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2022-11-29 23:18:54.000000000","tz":-360},"subject":"mypy, Ceph: correct types of file-likes in backup and restore","message":"mypy, Ceph: correct types of file-likes in backup and restore\n\nThe backup() and restore() are invoked from methods of class\nBackupManager. Those methods call open() and pass these handles\ndown to the backup driver. Although these volumes correspond\nto the driver, they are not wrapped using the driver.\nThus, the type passed is not driver\u0027s wrapper.\n\nChange-Id: Ie58db6f70aac0b13fbe832ecee177146e5dda397\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6eedb9d54b71ee0bfdda56e44c3f95571185c7ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6eedb9d54b71ee0bfdda56e44c3f95571185c7ba"}]},"branch":"refs/heads/master"},"7c163adce6b572350f871ba08ff508f9195d7b6f":{"kind":"REWORK","_number":2,"created":"2022-12-02 00:55:55.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/93/866093/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/866093/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/866093/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/866093/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/866093/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/866093/2"}}},"commit":{"parents":[{"commit":"7c1a5ce7b11964da4537fd6a7d157ede646b9e94","subject":"Merge \"nit: correct storwize_portset help text\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7c1a5ce7b11964da4537fd6a7d157ede646b9e94"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2022-11-29 23:18:54.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2022-12-02 00:55:06.000000000","tz":-360},"subject":"mypy, Ceph: correct types of file-likes in backup and restore","message":"mypy, Ceph: correct types of file-likes in backup and restore\n\nThe backup() and restore() are invoked from methods of class\nBackupManager. Those methods call open() and pass these handles\ndown to the backup driver. Although these volumes correspond\nto the driver, they are not wrapped using the driver.\nThus, the type passed is not driver\u0027s wrapper.\n\nChange-Id: Ie58db6f70aac0b13fbe832ecee177146e5dda397\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7c163adce6b572350f871ba08ff508f9195d7b6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7c163adce6b572350f871ba08ff508f9195d7b6f"}]},"branch":"refs/heads/master"},"68ef026ebe5dad25568a3f3dca804f880e2d40f4":{"kind":"REWORK","_number":3,"created":"2024-07-26 19:37:34.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/93/866093/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/93/866093/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/866093/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/866093/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/93/866093/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/93/866093/3"}}},"commit":{"parents":[{"commit":"d2e1b64681eb8a386c8e62b6f31831a2514657d4","subject":"Merge \"Hitachi: Fix to set correct object ID for LDEV nickname\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d2e1b64681eb8a386c8e62b6f31831a2514657d4"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2022-11-29 23:18:54.000000000","tz":-360},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2024-07-26 19:37:13.000000000","tz":-300},"subject":"DNM mypy, Ceph: correct types of file-likes in backup and restore","message":"DNM mypy, Ceph: correct types of file-likes in backup and restore\n\n[This is waiting for py38 to disappear.]\n\nThe backup() and restore() are invoked from methods of class\nBackupManager. Those methods call open() and pass these handles\ndown to the backup driver. Although these volumes correspond\nto the driver, they are not wrapped using the driver.\nThus, the type passed is not driver\u0027s wrapper.\n\nChange-Id: Ie58db6f70aac0b13fbe832ecee177146e5dda397\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68ef026ebe5dad25568a3f3dca804f880e2d40f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/68ef026ebe5dad25568a3f3dca804f880e2d40f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
