)]}'
{"id":"openstack%2Fcinder~867824","triplet_id":"openstack%2Fcinder~master~I5c1a4cd0ba0394849987203888b942169550672e","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2022-12-16 15:04:58.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-02-28 22:43:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5c1a4cd0ba0394849987203888b942169550672e","subject":"Tests: Fix double mock of Popen (test_backup_ceph)","status":"MERGED","created":"2022-12-15 15:52:05.000000000","updated":"2023-02-28 22:46:12.000000000","submitted":"2023-02-28 22:43:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867824","meta_rev_id":"763731903491c460eb1c31b5c92737e7b738ab48","_number":867824,"virtual_id_number":867824,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-28 22:43:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2023-02-28 20:01:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2023-01-19 17:15:38.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-10 18:19:42.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2023-02-28 20:01:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"}],"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-15 15:52:16.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-12-15 15:57:17.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-12-15 16:06:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-15 16:46:35.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2022-12-15 17:16:54.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-15 17:17:08.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-12-15 17:30:43.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-15 19:19:38.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-16 16:36:50.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2022-12-16 17:06:41.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-12-17 19:00:06.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2022-12-18 12:32:25.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-19 11:37:40.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-01-10 15:39:48.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2023-01-18 16:07:39.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-01-18 16:31:00.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-01-18 17:15:12.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-19 17:15:38.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2023-01-22 10:37:16.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-28 20:01:01.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"d7b1bedd0c5032a97f78927b96f7748ed20fbfa8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-12-15 15:52:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df31d7800865c6b6847d326dad97c17228642a5e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-12-15 15:52:16.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1860 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"bd3e996614ad4fe599697bbe6113d79c1bf3a6af","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-15 15:57:17.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"f09d393c2ed6ce558019d3cc84a220d1bcce82d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 16:06:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/493aea08561b4d3f9984db71385df7de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f234ed9a914d4a2f8030e4d311a5c82b : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3e66f73be654e3eb637a791e418d870 : SUCCESS in 10m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6740ce782e9c48d4a991fe1225cdb66d : SUCCESS in 10m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c7249563404bf8b1a5f6bc63396ee0 : SUCCESS in 11m 25s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/21f8f75f7083463d99550b12e169b56c : SUCCESS in 13m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/de68d2ecea48491a8e8a26c25b342521 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"fcca8693d6e2b6978bc152fc73a5d4851755ac15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 16:29:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30a6e3e7f1904124867114e9d6eb9138\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf98035451914589a8d0432edca7ef38 : FAILURE in 34m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/58b2b78e99d94022bed64bacc06b2187 : FAILURE in 32m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb112c0a5fbe9113d2afb24df22a874455da3ea9","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-12-15 16:46:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"38fef0c9ce05e2113f4873699bc4304441ec7d78","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-12-15 17:16:54.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/24/867824/1/check/DellEMC_PowerFlex/dbc5aad/DellEMC_PowerFlex/1486 : SUCCESS in 1h 09m 40s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"d75af7cfd03145fb996f6ae74a1fa446d9f879ad","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-12-15 17:17:08.000000000","message":"Patch Set 1:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/868","accounts_in_message":[],"_revision_number":1},{"id":"f0b1e3f46e2c421af407f58a942bcb39043fa1ca","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-15 17:30:43.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/867824 : SUCCESS in 1h 34m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d3ec0e43c7f11ba405a11e1c6aae3c10563eefc5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-15 17:35:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/867824 : SUCCESS in 1h 41m 13s","accounts_in_message":[],"_revision_number":1},{"id":"87985963dbaa8d7704da082c7b6ee09d29e765fd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-15 17:38:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/24/867824/1/check/iscsi-driver-alletra-6k/bfdb796/bfdb796.html : FAILURE in 1h 44m 21s","accounts_in_message":[],"_revision_number":1},{"id":"d4593b1538650c625385036ac27efd94c5222633","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-12-15 19:19:38.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-24-867824-1/ns5_nfs : SUCCESS in 3h 26m 37s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-24-867824-1/ns5_iscsi : SUCCESS in 3h 27m 3s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"cc46c19ff6a08282fb54b4c89d640bffd711a4b9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-12-16 11:57:10.000000000","message":"Patch Set 1:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/872","accounts_in_message":[],"_revision_number":1},{"id":"311dedc35b06e76d7a3a8568628a8566007f7cb7","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-12-16 14:50:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9e6d91e4ee85c16c9c312216a87d25e77685353","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-12-16 14:50:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"50fdd16d4ea05ba9834c7975c0f1599d5bd3caff","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-12-16 14:50:28.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1866 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"72d7821c3ff6a4e46db0ff7724c3d37afa77c44d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-16 14:53:47.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"dc9cbfef10907f0f13e85f3443b240120b3ed66e","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-12-16 15:04:58.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"445bf04380d0083697f61bb56da96610f3690051","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 15:10:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c528700e1e243e5a09805786a705ba8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ce232e82c56444496ec14ef36ebea16 : SUCCESS in 17m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2ccaf4cfea64454d81943a1fd56c6d7c : SUCCESS in 18m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cd0cac584dba443c0cb7dcb9f86ea2c1f1ab106e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 15:13:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81f14504c74543d2b834007ebac52af6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6969aa908c7b4a1a9ff193acd6ab3ad1 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0f01649b6df4485ac45adf9e1ce933f : SUCCESS in 10m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb5e89f1f9a44f87a4e8713557b0cb90 : SUCCESS in 10m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bce82d39c7842a7b975712ec3e7f40b : SUCCESS in 14m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c0edbbb6d2174b1289502630ee204497 : SUCCESS in 14m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3211e6bbe28c48918f15f09baccef444 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":2},{"id":"fe858c9cffebf47368e7dee523414094bbdce4ef","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-16 16:00:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/24/867824/2/check/iscsi-driver-alletra-6k/1def896/1def896.html : FAILURE in 1h 06m 38s","accounts_in_message":[],"_revision_number":2},{"id":"5c33b7698139b8d30e36400271b2f127db829350","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-16 16:24:02.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/867824 : SUCCESS in 1h 29m 38s","accounts_in_message":[],"_revision_number":2},{"id":"28d86643eb1fe4c8e4f33664372b59f7efcda983","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-16 16:27:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/867824 : SUCCESS in 1h 30m 43s","accounts_in_message":[],"_revision_number":2},{"id":"33df810c58ce49f6e847726d4d89279207bd7530","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2022-12-16 16:36:50.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/24/867824/2/check/DellEMC_SC_FC/03f8645/DellEMC_SC_FC/1113 : SUCCESS in 1h 06m 59s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/24/867824/2/check/DellEMC_SC_iSCSI/8441bbc/DellEMC_SC_iSCSI/1097 : SUCCESS in 1h 22m 50s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"0a59189fb97ba89d23a23d54955fbce23e2a1f32","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-12-16 16:37:57.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/24/867824/2/check/DellEMC_PowerFlex/1e2f501/DellEMC_PowerFlex/1492 : FAILURE in 1h 23m 58s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"e7fda1ef2fe2e431a141ba5b817346a3b0746663","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-16 17:06:41.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/24/867824/2/check/DellEMC_Unity_FC/f927e86/DellEMC_Unity_FC/2088 : FAILURE in 1h 33m 07s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/24/867824/2/check/DellEMC_Unity_iSCSI/cf1cdd2/DellEMC_Unity_iSCSI/1944 : FAILURE in 2h 16m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"08c790d4bc34372bca2d7278a92f8c4eb5af0318","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-12-16 18:18:23.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-24-867824-2/ns5_nfs : SUCCESS in 3h 25m 19s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-24-867824-2/ns5_iscsi : SUCCESS in 3h 27m 35s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"57cbca236f078be06ffe26934e89ebf75abf4c4d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-12-17 00:25:18.000000000","message":"Patch Set 2:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/876","accounts_in_message":[],"_revision_number":2},{"id":"c9140daa6b25db976ae6581a347e3a32d716919c","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2022-12-17 19:00:06.000000000","message":"Patch Set 2:\n\n* hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/24/867824/2/fc : FAILURE in 2h 03m 07s\n* hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/24/867824/2/iscsi : SUCCESS in 1h 31m 16s","accounts_in_message":[],"_revision_number":2},{"id":"fdfeeba0529143dcaaef4704d7b9da86b63576d0","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-12-18 12:32:25.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/24/867824/2/20221218175723/NEC-FC : SUCCESS in 3h 34m 02s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/24/867824/2/20221218175723/NEC-ISCSI : SUCCESS in 3h 34m 59s","accounts_in_message":[],"_revision_number":2},{"id":"1ee36fb00d8edb7fce02a96427f7018ec39d348f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-12-19 11:37:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/24/867824/2/20221219134354/NEC-V-FC : SUCCESS in 5h 38m 58s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/24/867824/2/20221219134354/NEC-V-ISCSI : SUCCESS in 6h 53m 42s","accounts_in_message":[],"_revision_number":2},{"id":"1b77ee06ec0d913a22fded03b5c7165338f7b630","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-01-06 15:37:18.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5d1abe1634d8dde9086a88f090324124600e531a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 15:45:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa814f20987e45d3a47c36e60eb0f300\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c6b227c5d4e4dd9b52dc0c7b8fae95e : FAILURE in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d6e3e80fbca459f9f79d520a9a33618 : FAILURE in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/570f768a52964e7d96b8667b9ab998dc : FAILURE in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bee5f5a3dba04d849641963804e54f21 : FAILURE in 4m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c078f1a97b65421b8b58f84ff2a3e3ec : FAILURE in 2m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/656f074620e44e40a6d55f9b3b52d07c : FAILURE in 2m 51s","accounts_in_message":[],"_revision_number":3},{"id":"e29d9efdd348a53b839ebfa21d738292913df962","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-06 16:43:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77f8ce90c94242cc9636adfc3cdc57e6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/17c8ca5cd658450bb7ab80e9ab2f56f7 : SUCCESS in 23m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1580ba6ed80d485caaa46c094fd1bfeb : SUCCESS in 20m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a83baeda89f6066d8b05a8384f5f9857e1cf2bfa","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-06 16:52:48.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/24/867824/3/check/iscsi-driver-nimble/b295bb6/b295bb6.html : FAILURE in 1h 14m 01s","accounts_in_message":[],"_revision_number":3},{"id":"461ebaf9c0aa9217053f6aa816e2d89d67461295","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-06 16:55:59.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/24/867824/3/check/iscsi-driver-alletra-6k/480d186/480d186.html : FAILURE in 1h 17m 37s","accounts_in_message":[],"_revision_number":3},{"id":"6713615304853db52602b933c8a2f62fefd7bd76","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-06 17:09:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/867824 : SUCCESS in 1h 29m 21s","accounts_in_message":[],"_revision_number":3},{"id":"0d28540886b2fa66c28feba4ba7281ab8827cc57","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-06 17:15:58.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/24/867824/3/check/DellEMC_Unity_FC/d5d110c/DellEMC_Unity_FC/2114 : SUCCESS in 1h 38m 18s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/24/867824/3/check/DellEMC_Unity_iSCSI/6b6f088/DellEMC_Unity_iSCSI/1985 : FAILURE in 6s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"8c5444e7bb1b45c5b374a02def97928ddd17976e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-06 17:18:07.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/867824 : SUCCESS in 1h 38m 55s","accounts_in_message":[],"_revision_number":3},{"id":"5eb7233d076540193edd9521faf136eba5b9e812","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-01-07 23:05:33.000000000","message":"Patch Set 3:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/24/867824/3/20230108022024/NEC-V-ISCSI : SUCCESS in 2h 59m 55s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/24/867824/3/20230108022024/NEC-V-FC : FAILURE in 5h 45m 06s","accounts_in_message":[],"_revision_number":3},{"id":"0c5826e67582a3fd39acee88a233455462524ba1","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-01-11 02:36:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/24/867824/3/20230111080233/NEC-FC : SUCCESS in 3h 33m 29s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/24/867824/3/20230111080233/NEC-ISCSI : SUCCESS in 3h 34m 08s","accounts_in_message":[],"_revision_number":3},{"id":"14070e939c99441048ddfb94781377b225950783","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-01-18 16:06:59.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"edfbd93b72a4077d15f6fa15dcce592e85870532","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-18 16:07:39.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"4b06be390c66957d1a72b930f63f17db51ed893d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 16:22:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0975e1f782c34f79af215be060644f96\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87a36ed2c4c04879a6024f371959c170 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e5242f2f0e64d46817ddac6a2a354de : SUCCESS in 12m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/874024b0ad034292a78dc540433cb804 : SUCCESS in 12m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdb2cd779e9c475cbeeb93b8ec15d846 : SUCCESS in 11m 46s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c19647cfe225433fb47a7fd0ba788057 : SUCCESS in 12m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/95fd367d714d4cfc9c96bbe0b66503a2 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":4},{"id":"7257d72efe855e77f7344416512f8838670aa9f0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-18 16:22:49.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/24/867824/4/check/iscsi-driver-nimble/d68ca95/d68ca95.html : FAILURE in 14m 41s","accounts_in_message":[],"_revision_number":4},{"id":"913b668f9a128cd98a94b9bbfadd8efc06396d41","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-01-18 16:31:00.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/18_01_2023_16_22_38\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"2c49025e959823e48d0bc3be49861c9008ce44a6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-18 17:14:28.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/24/867824/4/check/iscsi-driver-alletra-6k/e18ca06/e18ca06.html : FAILURE in 26m 06s","accounts_in_message":[],"_revision_number":4},{"id":"995c194f3d82b46d35167904d9d5f0f346a2e023","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-18 17:15:12.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2c1598a638d44668920728d5223527d5 : SUCCESS in 1h 06m 06s","accounts_in_message":[],"_revision_number":4},{"id":"e14ca834363ae1eca82c1e1a20f31934c618e0e8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-18 17:19:03.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/867824 : FAILURE in 25m 56s","accounts_in_message":[],"_revision_number":4},{"id":"e22a9f0fad53dcc4c15fbd445cf4950340d42f29","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-18 17:37:33.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/867824 : FAILURE in 23m 30s","accounts_in_message":[],"_revision_number":4},{"id":"b3979064f9f13197f4c3a7ce7085ca99d745fd6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 18:07:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd9122ad5873436ebfd134ef149fde61\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8299d769404f4632adc195ba226487ca : FAILURE in 32m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/28fef493e4284e6cb328bd6c4249087a : FAILURE in 33m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f93379299032a297e0d4ce2dbb8f4d23cf877b50","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-01-19 00:08:15.000000000","message":"Patch Set 4:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/960","accounts_in_message":[],"_revision_number":4},{"id":"9ffed7b7ac2a9b2134b4808f7237d0e35ceca2c6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-19 01:11:33.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-24-867824-4/ns5_nfs : FAILURE in 8h 42m 8s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-24-867824-4/ns5_iscsi : FAILURE in 9h 4m 6s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"4bb433c432dabb9c1696db6a5d7b873af8c74ab3","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2023-01-19 17:15:38.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"52f07bea61b2de6ba8198cb3a918935512bcc5b2","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-01-21 11:24:35.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/24/867824/4/20230121144005/NEC-V-ISCSI : SUCCESS in 3h 15m 34s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/24/867824/4/20230121144005/NEC-V-FC : FAILURE in 5h 44m 26s","accounts_in_message":[],"_revision_number":4},{"id":"5383e633ebac920808492fffcda6083aa0eab69b","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-22 10:37:16.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/24/867824/4/ : FAILURE in 56m 25s","accounts_in_message":[],"_revision_number":4},{"id":"c82b67871d098bd48883a7c182b9b8cd06ed22e9","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-02-10 18:19:42.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3f99a3e968c88c164a2400cf36a18551929b4196","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-02-28 20:01:01.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e20e2bdbb3300a7323d4970c1c8e6ec6eafcdddb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 20:01:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7bccf73fab8c3f539c8c2dcf42e946d41786a743","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 22:43:47.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd6ecd25622a44c3813bae23d759d149\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/704f84852b484faab0cae6ddd02f5e8b : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27dda34b963f47ac8873ec4734bac258 : SUCCESS in 13m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/beabcd1ce51b49adbf4f4b7f35151d10 : SUCCESS in 15m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40524f7d2aaa48549158d5ee71b123b9 : SUCCESS in 13m 20s","accounts_in_message":[],"_revision_number":4},{"id":"3ed3fde35bcc2e9fa12647985c700f4a4beaf845","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 22:43:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"763731903491c460eb1c31b5c92737e7b738ab48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 22:46:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4052433fee8143d0ac7f7c33f2033d29\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20c6392a27424e39a2f928ca7a8878f5 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8d9e0bf976d9d367e203a6ba9a0ddd82b73d6b20","revisions":{"92b06c0ce3cc5c2224a91d0628d883a65e11c4d3":{"kind":"REWORK","_number":1,"created":"2022-12-15 15:52:05.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/24/867824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/867824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/867824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/867824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/867824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/867824/1"}}},"commit":{"parents":[{"commit":"24a783408879abd479e77dd88b43bf2a1be048fe","subject":"Merge \"JovianDSS: extend option description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/24a783408879abd479e77dd88b43bf2a1be048fe"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-12-15 15:47:14.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-12-15 15:49:42.000000000","tz":-300},"subject":"Tests: Fix double mock of Popen (test_backup_ceph)","message":"Tests: Fix double mock of Popen (test_backup_ceph)\n\nThis is an error in Python 3.11\u0027s mock.  Stop mocking\nPopen when it is already mocked by @common_mocks.\n\nChange-Id: I5c1a4cd0ba0394849987203888b942169550672e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/92b06c0ce3cc5c2224a91d0628d883a65e11c4d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/92b06c0ce3cc5c2224a91d0628d883a65e11c4d3"}]},"branch":"refs/heads/master"},"66aa952321d7c34c78b636df04610cb850f2334c":{"kind":"REWORK","_number":2,"created":"2022-12-16 14:50:18.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/24/867824/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/867824/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/867824/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/867824/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/867824/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/867824/2"}}},"commit":{"parents":[{"commit":"24a783408879abd479e77dd88b43bf2a1be048fe","subject":"Merge \"JovianDSS: extend option description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/24a783408879abd479e77dd88b43bf2a1be048fe"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-12-15 15:47:14.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-12-16 14:49:55.000000000","tz":-300},"subject":"Tests: Fix double mock of Popen (test_backup_ceph)","message":"Tests: Fix double mock of Popen (test_backup_ceph)\n\nThis is an error in Python 3.11\u0027s mock.  Stop mocking\nPopen when it is already mocked by @common_mocks.\n\nChange-Id: I5c1a4cd0ba0394849987203888b942169550672e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/66aa952321d7c34c78b636df04610cb850f2334c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/66aa952321d7c34c78b636df04610cb850f2334c"}]},"branch":"refs/heads/master"},"73e345b73c3064a60bd961d89565d8c1343e3b63":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-01-06 15:37:18.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/24/867824/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/867824/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/867824/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/867824/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/867824/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/867824/3"}}},"commit":{"parents":[{"commit":"24a783408879abd479e77dd88b43bf2a1be048fe","subject":"Merge \"JovianDSS: extend option description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/24a783408879abd479e77dd88b43bf2a1be048fe"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-12-15 15:47:14.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-01-06 15:37:18.000000000","tz":0},"subject":"Tests: Fix double mock of Popen (test_backup_ceph)","message":"Tests: Fix double mock of Popen (test_backup_ceph)\n\nThis is an error in Python 3.11\u0027s mock.  Stop mocking\nPopen when it is already mocked by @common_mocks.\n\nRelated-Bug: #2000436\nChange-Id: I5c1a4cd0ba0394849987203888b942169550672e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/73e345b73c3064a60bd961d89565d8c1343e3b63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/73e345b73c3064a60bd961d89565d8c1343e3b63"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8d9e0bf976d9d367e203a6ba9a0ddd82b73d6b20":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-01-18 16:06:59.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/24/867824/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/24/867824/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/867824/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/867824/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/24/867824/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/24/867824/4"}}},"commit":{"parents":[{"commit":"bc9910f0409c60d82788b9ad248327b377439c3a","subject":"Merge \"LVM-nvmet: Use nvmetcli as library instead of CLI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bc9910f0409c60d82788b9ad248327b377439c3a"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-12-15 15:47:14.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-01-18 16:06:59.000000000","tz":0},"subject":"Tests: Fix double mock of Popen (test_backup_ceph)","message":"Tests: Fix double mock of Popen (test_backup_ceph)\n\nThis is an error in Python 3.11\u0027s mock.  Stop mocking\nPopen when it is already mocked by @common_mocks.\n\nRelated-Bug: #2000436\nChange-Id: I5c1a4cd0ba0394849987203888b942169550672e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d9e0bf976d9d367e203a6ba9a0ddd82b73d6b20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d9e0bf976d9d367e203a6ba9a0ddd82b73d6b20"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
