)]}'
{"id":"openstack%2Fcinder~868120","triplet_id":"openstack%2Fcinder~master~I04a2b070c88e5d23fc79fbe5651e1b64a1d4db2f","project":"openstack/cinder","branch":"master","topic":"mypy","attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2022-12-19 18:05:41.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I04a2b070c88e5d23fc79fbe5651e1b64a1d4db2f","subject":"WIP: mypy: cinder/api/v3/volumes.py","status":"NEW","created":"2022-12-19 18:05:34.000000000","updated":"2023-01-15 21:28:35.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"47c8ebf6775476f0a740f1504e44f731fa3134f4","_number":868120,"virtual_id_number":868120,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-12-20 17:45:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-19 18:05:41.000000000","updated_by":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"reviewer":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"state":"CC"},{"updated":"2022-12-19 18:06:14.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2022-12-19 18:32:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-19 18:34:07.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-19 19:08:59.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-19 20:19:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-20 00:28:54.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2022-12-20 08:23:39.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-20 16:10:26.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-22 01:04:27.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"}],"messages":[{"id":"ec3159b69ae9153999dc6f2799379b1f309395d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-12-19 18:05:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d8de994897371df6b478593f095e47a4eea2623","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-12-19 18:05:41.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1874 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"2d60ea63d0f6570ec66713e43f130e1f94217ab2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-19 18:06:14.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"1381df3a459d985987d5549ec0708b0cd2bf0c77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 18:32:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a1a6dc76a4848979204fb5bdb0bd8f6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba8d58718c114b65839e1b7cf13a6022 : FAILURE in 13m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4cbd7d1fc2da44bba1ea683f045aa0ba : SUCCESS in 23m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2376148b71ced547e1bda80224053853dd259d0c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-12-19 18:34:07.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-20-868120-1/ns5_nfs : FAILURE in 28m 3s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-20-868120-1/ns5_iscsi : FAILURE in 28m 3s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"b733899b5528e39a7d901dfa9b818117e64befb0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-19 19:08:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/868120 : FAILURE in 1h 00m 59s","accounts_in_message":[],"_revision_number":1},{"id":"75ea6e67ef2cd202a7000d330f1c8ed88ddf42d9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-19 19:20:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/20/868120/1/check/iscsi-driver-nimble/e318e7f/e318e7f.html : FAILURE in 1h 09m 02s","accounts_in_message":[],"_revision_number":1},{"id":"1f3e64bc3a3e4cce4d9195c6b6be088b99449fcb","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-19 19:29:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/868120 : FAILURE in 1h 21m 21s","accounts_in_message":[],"_revision_number":1},{"id":"13c4e96626bba41b73a678ef9d67a8d5fb4bf2c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 20:19:34.000000000","message":"Patch Set 1: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/554ccc933f36450f8ec6b8f1306334b9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23fc6280ed2f42a8aec7612515bd7a25 : FAILURE in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e740f920b0b04efda047906b3cf35cc5 : FAILURE in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd72da1a31cc449f8db0bdb81eaf3486 : SUCCESS in 9m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ef212a9563b4196813f9a291cf82179 : SUCCESS in 11m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/0634fc124d614f46a27887875856b977 : FAILURE in 39m 58s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/efd71b47fd2a49bbb0ea9f614b0ef9cc : FAILURE in 52m 15s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/93f58a26dc614fa491c1ad18e343bcab : SUCCESS in 1h 10m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/850777071e3e491bb39bf85462f9f9f3 : SUCCESS in 47m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0fcc22df0bff40308fbcf8e7a690dae6 : SUCCESS in 15m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1caa91814f064f389092b052e03dc15b : SUCCESS in 5m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b6fb6c3aaea1469280587e2a5f190f45 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4cade17f5b2c4fb497ace4af6368c366 : FAILURE in 4m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bafe5773f8eb4243b5d4ae52f4b4cb0a : SUCCESS in 5m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/60495e484c1d402380aec6fb08b805bc : SUCCESS in 37m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a894e0501bf44e06870e9c8d0f6fa7e7 : FAILURE in 5m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7e549e0b304c4c91a096cf3aaf798848 : FAILURE in 34m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/440bc0cabfbd48439eddbad32782b07d : FAILURE in 36m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8ac2ea37fcfa44878cc5a571b6a2e3f6 : SUCCESS in 1h 39m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/efdf0b5a8d074b4c955842a02426b3f2 : FAILURE in 56m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/80be127b37ec4bfea9c43dd5b3069b51 : SUCCESS in 1h 36m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/62d378a20f4e498ea18065de19008efc : SUCCESS in 29m 47s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/731358616a37422d82156b487fc5dbe5 : SUCCESS in 1h 29m 43s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e2574d12a04c4f0ab87832f310422069 : SUCCESS in 2h 08m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29ad706f9839435891f2a3de1733d254 : SUCCESS in 49m 01s","accounts_in_message":[],"_revision_number":1},{"id":"0d95abb3806c4076133f1b7f16f1aff8b747840e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-12-20 00:28:54.000000000","message":"Patch Set 1:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/889","accounts_in_message":[],"_revision_number":1},{"id":"e7ce66779b2701997761b1b78830c86d1801ef32","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2022-12-20 08:23:39.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/20/868120/1/20221220163241/NEC-ISCSI : FAILURE in 49m 05s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/20/868120/1/20221220163241/NEC-FC : FAILURE in 50m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e2ec35f08005205cb1468d7f422007a9b759dc2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2022-12-20 15:18:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5190a5160e1426eafe7926646f73d4fb407d8639","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34588,"name":"DataCore Software","email":"cinder-ci@datacore.com","username":"Arun"},"date":"2022-12-20 15:18:18.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nLog path: https://github.com/arun-kv/cinder-tempest-logs/tree/main/#1883 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"8b4d7ace2ecb2a979464de6170560a75df3a28f0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-20 15:18:49.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"a3cb1006c08183c1b007c77619b355421b369e98","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2022-12-20 15:40:40.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : MERGER_FAILURE in 1m 26s","accounts_in_message":[],"_revision_number":2},{"id":"b9731acdb3313ac16aa5c26d0920ec2327b85ff8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-12-20 15:44:34.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-20-868120-2/ns5_nfs : FAILURE in 25m 57s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-20-868120-2/ns5_iscsi : FAILURE in 25m 52s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"9e590c976fc853475be9efd8a82b3a5ce7f65285","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 15:58:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed820fd609ec4951afbf70c986d85654\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4fb2c5c40a7446efa610c0599084ed25 : FAILURE in 14m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/708ec1538bc544db80ddcfb2c042136d : FAILURE in 38m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa5789f527a0b86d464072513f09e26bea1ec11e","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2022-12-20 16:10:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/20/868120/2/check/fujitsu-eternus-dx-iscsi/b7278b8 : FAILURE in 51m 28s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/20/868120/2/check/fujitsu-eternus-dx-fc/a1f57b0 : FAILURE in 51m 06s","accounts_in_message":[],"_revision_number":2},{"id":"5fb8c02ff7a64164ea86240bc25294f198f9f02c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-20 16:20:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/868120 : FAILURE in 1h 00m 33s","accounts_in_message":[],"_revision_number":2},{"id":"61e68e9bd19de999399511be56e3b3b866d604c7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2022-12-20 16:26:04.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/868120 : FAILURE in 1h 05m 14s","accounts_in_message":[],"_revision_number":2},{"id":"890679175858c20e4b399b578606097fd7cb9dc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 17:45:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/093f5c99cfed49ee9c076919a52098e0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeff8c297c66408a82d993b3208205a3 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39942813a39143d49c2a413e1a7b61b9 : FAILURE in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b32e8306589b40e984b0b4e75b625848 : SUCCESS in 9m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57147f54ebf049dfb69278ba1df5adbb : SUCCESS in 13m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/5daf731742ea4b19a2cf4d4b6c9a5fb7 : FAILURE in 55m 37s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ff3bbe531b1e4abcaf6e97b4ca335563 : FAILURE in 54m 21s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/380b9c60d30a486ea40de49e7b2fcee4 : SUCCESS in 1h 15m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a8b0d2a741d43118f1fe46324f59435 : SUCCESS in 1h 01m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/35244e48ddbd42e0977a08132aff47f5 : SUCCESS in 17m 38s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8de2c913380f4044b0a91a898ae06e77 : SUCCESS in 5m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/57c69109a8424cdb963a642c0a6ad111 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6ad40eed43ce488488943110aa230aeb : FAILURE in 5m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8174a189eaf64ffba0611409b04c1ff9 : SUCCESS in 5m 36s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dc8c71b58fac4962b2dc096aa7817f86 : SUCCESS in 45m 10s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0f0f38dfff3d47b7941079ab06ae7feb : FAILURE in 6m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d8b41be4db6743629dcd1662a8b45c46 : FAILURE in 30m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7303494175b94a85b43a4b58ccb796ce : FAILURE in 26m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4c9a182619c04c46bd04f8087c81fb0d : SUCCESS in 2h 05m 08s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/33c665f3a8b2489b934dde35467a5150 : FAILURE in 1h 05m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/46189ec8aad247f3b9413bfb52aea139 : FAILURE in 1h 36m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/1769f3eb75ad43c2b6fc20c21c4e91cd : SUCCESS in 31m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/13ebc9975c9241c6902c55c6453939ec : SUCCESS in 1h 30m 56s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/08144d77b7b940218c3dd951dd849dc5 : SUCCESS in 2h 21m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e9874295ad046859c22b38372befff3 : SUCCESS in 52m 05s","accounts_in_message":[],"_revision_number":2},{"id":"6656fd270a747c05b3b3aadd6e41835c5f3f4657","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2022-12-20 19:49:11.000000000","message":"Patch Set 2:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/894","accounts_in_message":[],"_revision_number":2},{"id":"c45d39829af80d157e5a52a899a74cc5cdfb764a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2022-12-22 01:04:27.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/20/868120/2/20221222091055/NEC-V-ISCSI : FAILURE in 49m 43s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/20/868120/2/20221222091055/NEC-V-FC : FAILURE in 53m 29s","accounts_in_message":[],"_revision_number":2},{"id":"47c8ebf6775476f0a740f1504e44f731fa3134f4","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-01-15 21:28:35.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/20/868120/2/20230116053718/NEC-ISCSI : FAILURE in 49m 23s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/20/868120/2/20230116053718/NEC-FC : FAILURE in 51m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"641d9a4283d50f1646eade9a1d500421bb8f7dbe","revisions":{"b86010f7fef532a582a01526acb1ac037b21ad4e":{"kind":"REWORK","_number":1,"created":"2022-12-19 18:05:34.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/20/868120/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/868120/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/868120/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/868120/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/868120/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/868120/1"}}},"commit":{"parents":[{"commit":"94c4c9d526d07f3587b318d84356b056dd94e42c","subject":"mypy: cinder/volume/flows/api/manage_existing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/94c4c9d526d07f3587b318d84356b056dd94e42c"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-12-19 15:07:10.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-12-19 18:04:35.000000000","tz":-300},"subject":"WIP: mypy: cinder/api/v3/volumes.py","message":"WIP: mypy: cinder/api/v3/volumes.py\n\nChange-Id: I04a2b070c88e5d23fc79fbe5651e1b64a1d4db2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b86010f7fef532a582a01526acb1ac037b21ad4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b86010f7fef532a582a01526acb1ac037b21ad4e"}]},"branch":"refs/heads/master"},"641d9a4283d50f1646eade9a1d500421bb8f7dbe":{"kind":"REWORK","_number":2,"created":"2022-12-20 15:18:06.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/20/868120/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/20/868120/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/868120/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/868120/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/20/868120/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/20/868120/2"}}},"commit":{"parents":[{"commit":"94c4c9d526d07f3587b318d84356b056dd94e42c","subject":"mypy: cinder/volume/flows/api/manage_existing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/94c4c9d526d07f3587b318d84356b056dd94e42c"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-12-19 15:07:10.000000000","tz":-300},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2022-12-20 15:17:54.000000000","tz":-300},"subject":"WIP: mypy: cinder/api/v3/volumes.py","message":"WIP: mypy: cinder/api/v3/volumes.py\n\nChange-Id: I04a2b070c88e5d23fc79fbe5651e1b64a1d4db2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/641d9a4283d50f1646eade9a1d500421bb8f7dbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/641d9a4283d50f1646eade9a1d500421bb8f7dbe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
