)]}'
{"id":"openstack%2Fcinder~869188","triplet_id":"openstack%2Fcinder~stable%2Fxena~I0ca55abf9975c5a3f9713ac5dd5be39083e04554","project":"openstack/cinder","branch":"stable/xena","topic":"tox4-pin-testing","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-01-06 18:05:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0ca55abf9975c5a3f9713ac5dd5be39083e04554","subject":"DNM: tetsing tox\u003c4 pin","status":"ABANDONED","created":"2023-01-04 18:21:08.000000000","updated":"2023-01-06 18:05:02.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":869187,"cherry_pick_of_patch_set":1,"meta_rev_id":"4a440c90b13dc31ca3bf0d53ebf8c7a5c8d0f019","_number":869188,"virtual_id_number":869188,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-01-05 22:31:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-04 18:31:37.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-01-04 19:13:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-04 20:33:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-04 20:46:55.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-01-04 20:47:00.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-04 20:53:36.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"0000ce332f22f2445f0baf6adbda4762c9a19832","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-04 18:21:08.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"e96a38a5e5b86ecb94e801d92d37df7619916ca5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-04 18:31:37.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"bded3362ce9a8b8db8352e9c8e1cd5d4cf78cdaa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 19:13:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/704c22c8ffc244f5baab432800db71d0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8cb8af2e40d0404790cc6fa71a118952 : SUCCESS in 16m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3cb6f9cd5a904ae78048def5ec04e51a : SUCCESS in 21m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bbda2b77544b9321ff02b16c68aea6fc50a5d724","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-04 19:37:05.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/88/869188/1/check/iscsi-driver-alletra-6k/3b680af/3b680af.html : FAILURE in 1h 07m 45s","accounts_in_message":[],"_revision_number":1},{"id":"d27a0afe9828f6183d18861c2e1a7aa2f1b03f34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 20:33:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1945fef949dd42eda6165b6c9763bd14\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56b446316786425aadccad843e9e2ecb : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e59f43b8ede4ba99f9929468f8fd407 : SUCCESS in 11m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5df772d34674c6ab0d4195403e1df8c : SUCCESS in 9m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/510a415b4c4c40a29c363f00aae779aa : SUCCESS in 11m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac1331ef6d8347a28372b75bc3c9d664 : SUCCESS in 10m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/a8ed7d35cb204978bc86627f4aac2c24 : SUCCESS in 1h 08m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7aefbd4820c74132addb768831d4e6eb : SUCCESS in 1h 22m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d9c37bc8c3449678beb71b9ed61340f : SUCCESS in 54m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/afe004830b4947f89d26f6eb85a0d58c : SUCCESS in 12m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/12e0c8d3cc644895b99e0af030ec65f7 : FAILURE in 4m 59s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/207fb50290e2487ab2623e65122cfdb3 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8a88573ac04441d4b2980c885e5c6f43 : SUCCESS in 5m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ab47566ac34f43fd8efd94f0b26c1dfb : SUCCESS in 5m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5dd509005dc44c41a2dc2f1a669f2f84 : SUCCESS in 45m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/48488d3e329a468faf1785de198de4f1 : SUCCESS in 4m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0e2bc051def349e48c7daf65b8506195 : SUCCESS in 1h 27m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bbfb437c48f14dc8b46906f5959e7b18 : FAILURE in 1h 26m 16s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2e0616ffc6d140b7b73ccd52788a954d : SUCCESS in 1h 31m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/51ae8143695247c2b0e0e8df17ee64dc : SUCCESS in 1h 14m 06s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fa341bc5c7dc435f81cf7e0da71967cb : SUCCESS in 1h 13m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5ed0d9a9bc314f6bb8a7643286757346 : SUCCESS in 41m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/af0772be279b4045acae97066b62acc4 : SUCCESS in 1h 32m 28s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ed57a2c80049470f9beeada4d099f3f9 : SUCCESS in 2h 01m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2737e5d4bce4bc2a2d483ffece5f02a : SUCCESS in 59m 01s\n\nWarning:\n  Comments left for invalid file cinder/volume/manager.py","accounts_in_message":[],"_revision_number":1},{"id":"1a8b82faedacc36fd7a7117f529ad1c6f723103b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-04 20:46:55.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"cd81882c98d55313952cfbc5006e57e6af1e5aca","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-04 20:47:00.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"62117d215e3d174eba07b293260f399ae708e297","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-04 20:47:03.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"7cb4ae01a7e17a4e5a903d1bff1a741972ebfef7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-04 20:51:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7382a3f6fe9e25170ecba166076691f9bd2110f0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-04 20:53:36.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"5a8d27230abef72bd5f976a69f5e18eefccd9f7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 21:23:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f381a0ce9c442608f7bedc4575b9844\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/02c2431ca86c448198bde65dbbeb6169 : SUCCESS in 16m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6897abda28a4da5ad1c8a5025b83a79 : SUCCESS in 20m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1122845e2bc566bfd2e1b488fdb87fbe6aad75dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 22:56:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/501edcb05ebb439a918faea7660e0e73\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28875748d01545f3862d0f078521db28 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30e4cd3350004cc08010464771f75a38 : SUCCESS in 10m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6efecdf5d3ec4f9aba26c34d4138d372 : SUCCESS in 9m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95f9fbb2009b480e83b971efb4c797c1 : SUCCESS in 10m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e25314db06544018bcb1573ad59a9908 : SUCCESS in 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/fe5824d7314f4ee09bc71e3721dd38eb : SUCCESS in 1h 06m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d4c50469eea940e48bb0c937760e8de5 : SUCCESS in 1h 17m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4afdc17cd3b244ada609e0e1ee45d424 : SUCCESS in 46m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c9e304cda6364377a1f0d6da6168c553 : SUCCESS in 13m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/13d2a8a83b2f43f9ae8fc06d33a4ace3 : FAILURE in 5m 03s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/32ae2090cb7e40d1b2f602d106ca05da : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3b98289dd8be48a78777d868f02eb4b4 : SUCCESS in 5m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/178ebdd5b24e4166891c6aa287351d05 : SUCCESS in 5m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8bfb65fdb1fa4cc5b7b3378891b9f80e : SUCCESS in 42m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b2294ea7eec54ebbb0c7c7dd24b06a63 : SUCCESS in 4m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f5bc58ad4e7c49a49c68c9ae85daad56 : SUCCESS in 1h 18m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9b1f0655514247259ba8eb70a8968c93 : FAILURE in 1h 20m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/de52913aa6d948269618b76e78b7ad45 : SUCCESS in 1h 45m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3c904af4f1e942e18da88c5622f881d7 : SUCCESS in 1h 14m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/da5e178cfaee444cb0effe693c5014c2 : SUCCESS in 1h 14m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/43d763b6a25f4ae89f82b5fe06fc2d02 : SUCCESS in 30m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/63bf6ac04940487fbedefd5e11a7b581 : SUCCESS in 59m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/577bc993ff624940a6eb16bad999ca5f : SUCCESS in 1h 56m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dae7978c5084426caaff98188da2ced6 : SUCCESS in 38m 31s\n\nWarning:\n  Comments left for invalid file cinder/volume/manager.py","accounts_in_message":[],"_revision_number":1},{"id":"801fc50d8f53c2aa58773cf889c0c2efe24bbd8b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-05 20:30:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4e2fb189ed88bfd6c7181bc30867ed1459c1105","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-05 20:36:58.000000000","message":"Topic set to tox4-pin-testing","accounts_in_message":[],"_revision_number":1},{"id":"d1dae22d0a1c072b580b22028f29267ec71314d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 20:57:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/616e7e5d29d64e4dad936d830a5952cb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dbc70493d42e4dd09f9c6bca30317e19 : SUCCESS in 20m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca3935b312484cdcb5738594aebf18cd : SUCCESS in 23m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5e25cb100c461a70dee63368c50e662139482f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 22:31:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3835d2af4964a73bacaac0794afa4af\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e569be526ef4a61a7e178278c51c54f : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52a13409d1954ecbb716401239795e75 : SUCCESS in 8m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bf89092938b4e288df5707ac7a8cc72 : SUCCESS in 8m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/779af73654e444ce971090e8773f6be3 : SUCCESS in 10m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c47ca66bd02a4c2dbb0f3d0202f91e1f : SUCCESS in 10m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/162d40001c604b01b6a3d36b5afdba88 : SUCCESS in 50m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2520051e5b494928a9993d9397daa011 : SUCCESS in 1h 00m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9381a157193e491785219ea9bfacc3eb : FAILURE in 27m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dd7826ef47cc4e09816abd166746fc4b : SUCCESS in 11m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a57e478da48c4b2db32739f9840311f6 : FAILURE in 4m 34s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/490962455dbf4c48aa353e7c28bc84ef : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ba637954d48f4271958d425d8493d47a : SUCCESS in 5m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a0e08556a1e64f3aadadf4b71f032a91 : SUCCESS in 5m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d8c75ec710e84dbb82fb9a5a0e41307d : SUCCESS in 42m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4ddb9a216ef847859ddabfa0c0246ed8 : SUCCESS in 4m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b8fd86e6c8f5401b989a5dfeeec4e1a9 : SUCCESS in 1h 10m 02s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bf5cbb3591c54b58b2444eee1f96ff32 : FAILURE in 1h 03m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/aa297847880243c8ac7865d77c8dc698 : SUCCESS in 1h 32m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c9ac222df7704dcf8ebf1944fc086eab : SUCCESS in 1h 15m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5e997b112ce6487a86011503f389d622 : SUCCESS in 1h 08m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c41b6cc8a6a9499291a1516ceeff2906 : SUCCESS in 37m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0fce77a8cb2a41cab27ee7477c183277 : SUCCESS in 1h 17m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/04a5f026a2234275bf1327db9663f6c7 : SUCCESS in 1h 43m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6e598a5bfaf483db9dda9df119324ae : SUCCESS in 46m 28s\n\nWarning:\n  Comments left for invalid file cinder/volume/manager.py","accounts_in_message":[],"_revision_number":1},{"id":"4a440c90b13dc31ca3bf0d53ebf8c7a5c8d0f019","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-06 18:05:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"867be5e54c380a95d3b598f5cf79cad8dc19a440","revisions":{"867be5e54c380a95d3b598f5cf79cad8dc19a440":{"kind":"REWORK","_number":1,"created":"2023-01-04 18:21:08.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/88/869188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/869188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/869188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/869188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/869188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/869188/1"}}},"commit":{"parents":[{"commit":"4c7dbde4401aba24ee58f526fc98976580d64e62","subject":"rbd: Fix snapshot delete when the source volume doesn\u0027t exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c7dbde4401aba24ee58f526fc98976580d64e62"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-04 18:20:00.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-04 18:21:08.000000000","tz":0},"subject":"DNM: tetsing tox\u003c4 pin","message":"DNM: tetsing tox\u003c4 pin\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/867849\nChange-Id: I0ca55abf9975c5a3f9713ac5dd5be39083e04554\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/867be5e54c380a95d3b598f5cf79cad8dc19a440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/867be5e54c380a95d3b598f5cf79cad8dc19a440"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
