)]}'
{"id":"openstack%2Fcinder~869305","triplet_id":"openstack%2Fcinder~master~I067bcd66cedfac392b63c1540fd5022de24626ac","project":"openstack/cinder","branch":"master","topic":"bug/1992160","attention_set":{},"removed_from_attention_set":{"37328":{"account":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"last_update":"2026-08-07 16:49:21.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-08-07 16:49:21.000000000","reason":"Change was submitted"},"30615":{"account":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"last_update":"2026-08-07 16:49:21.000000000","reason":"Change was submitted"},"35193":{"account":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"last_update":"2026-08-07 16:49:21.000000000","reason":"Change was submitted"},"36538":{"account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"last_update":"2025-12-19 10:14:03.000000000","reason":"\u003cGERRIT_ACCOUNT_36538\u003e replied on the change","reason_account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-08-07 16:49:21.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2026-08-05 14:00:41.000000000","reason":"\u003cGERRIT_ACCOUNT_36171\u003e replied on the change","reason_account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2026-08-07 16:49:21.000000000","reason":"Change was submitted"},"37470":{"account":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"last_update":"2026-08-07 16:49:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I067bcd66cedfac392b63c1540fd5022de24626ac","subject":"[SVf] lsportip needs to fetch IPs with \"host\" flag","status":"MERGED","created":"2023-01-10 06:39:14.000000000","updated":"2026-08-07 16:50:27.000000000","submitted":"2026-08-07 16:49:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":31,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869305-bug/1992160","meta_rev_id":"1dec07d718ad165641badd46a0bd1182b29edaf5","_number":869305,"virtual_id_number":869305,"owner":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-07 16:49:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":36348,"name":"Rahul Pawar","display_name":"Rahul Pawar","email":"rahul.p@ibm.com","username":"rahulp"},{"value":0,"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"all":[{"value":2,"date":"2026-07-15 00:19:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":1,"date":"2025-10-29 22:51:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},{"value":2,"date":"2026-08-07 14:57:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":1,"date":"2026-08-05 14:00:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":1,"date":"2025-08-29 04:35:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":1,"date":"2025-12-19 10:14:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":1,"date":"2025-08-28 12:11:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36348,"name":"Rahul Pawar","display_name":"Rahul Pawar","email":"rahul.p@ibm.com","username":"rahulp"},{"value":0,"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"all":[{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},{"value":1,"date":"2026-08-07 14:57:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":36348,"name":"Rahul Pawar","display_name":"Rahul Pawar","email":"rahul.p@ibm.com","username":"rahulp"},{"value":0,"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},{"value":0,"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"_account_id":36348,"name":"Rahul Pawar","display_name":"Rahul Pawar","email":"rahul.p@ibm.com","username":"rahulp"},{"value":0,"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32465,"name":"IBM Storage CI","email":"hchintha@in.ibm.com","username":"ibm-svf-user"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36348,"name":"Rahul Pawar","display_name":"Rahul Pawar","email":"rahul.p@ibm.com","username":"rahulp"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-10 06:40:00.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-10 07:12:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-10 08:10:50.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-01-10 08:45:24.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-10 09:00:33.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-01-10 09:09:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-10 09:21:32.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-01-10 10:06:45.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-10 12:25:55.000000000","updated_by":{"_account_id":32465,"name":"IBM Storage CI","email":"hchintha@in.ibm.com","username":"ibm-svf-user"},"reviewer":{"_account_id":32465,"name":"IBM Storage CI","email":"hchintha@in.ibm.com","username":"ibm-svf-user"},"state":"CC"},{"updated":"2023-01-10 12:29:04.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-01-12 07:19:27.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-12 07:39:16.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-01-12 08:18:15.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-12 09:56:57.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-19 15:13:17.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2023-01-20 05:49:27.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-01-20 07:26:57.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-23 17:09:41.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-01-23 18:06:51.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-01-23 19:04:26.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-01-24 08:16:50.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-01-24 10:53:44.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2023-01-24 10:57:47.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-01-25 13:18:22.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2023-02-15 08:35:41.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2024-02-13 06:32:07.000000000","updated_by":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"reviewer":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"state":"REVIEWER"},{"updated":"2025-07-22 14:13:16.000000000","updated_by":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"reviewer":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"state":"CC"},{"updated":"2025-07-22 14:18:32.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"CC"},{"updated":"2025-07-28 11:32:41.000000000","updated_by":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"reviewer":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"state":"CC"},{"updated":"2025-08-18 14:57:29.000000000","updated_by":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"reviewer":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"state":"REVIEWER"},{"updated":"2025-08-18 15:28:51.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-08-18 16:28:17.000000000","updated_by":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"reviewer":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"state":"CC"},{"updated":"2025-08-18 21:25:14.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2025-08-28 12:11:04.000000000","updated_by":{"_account_id":36348,"name":"Rahul Pawar","display_name":"Rahul Pawar","email":"rahul.p@ibm.com","username":"rahulp"},"reviewer":{"_account_id":36348,"name":"Rahul Pawar","display_name":"Rahul Pawar","email":"rahul.p@ibm.com","username":"rahulp"},"state":"REVIEWER"},{"updated":"2025-08-29 04:35:52.000000000","updated_by":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"reviewer":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"state":"REVIEWER"},{"updated":"2025-10-29 22:51:54.000000000","updated_by":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"reviewer":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"state":"REVIEWER"},{"updated":"2025-11-07 15:54:34.000000000","updated_by":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"reviewer":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"state":"REVIEWER"},{"updated":"2026-07-13 14:05:01.000000000","updated_by":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"reviewer":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"state":"CC"},{"updated":"2026-07-17 14:37:52.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"CC"},{"updated":"2026-07-17 14:38:14.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2026-08-07 14:57:40.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"}],"messages":[{"id":"d5fa2f16d22bdab9bc5b9e81f34396aa7d32b67e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"date":"2023-01-10 06:39:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4bb010dec927c78a306660e23bbb0bde4fad95b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-10 06:40:00.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8349d7d49d6e4459c9edce50de2532895c30418f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"date":"2023-01-10 06:50:23.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0545ace353674504a4081135b644bcca80110fa1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-10 06:51:10.000000000","message":"Patch Set 2:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":2},{"id":"88c2cb99ad7488c908d8b8d9c919342ea667251e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 07:12:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4541833601314adb9f05048bbcbf11a8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cfe160c46b994c83ae8cf5654e28641d : SUCCESS in 18m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9bc8284f832e48f09074ba52e71118d4 : SUCCESS in 19m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c66f88cc5066c9061f9c1654cbcf64168d15f683","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-10 08:10:50.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/05/869305/2/check/iscsi-driver-nimble/1d9272e/1d9272e.html : FAILURE in 1h 15m 09s","accounts_in_message":[],"_revision_number":2},{"id":"2330e91f0f1ddab5bb9e2e0b711e76b6a6fb0dcd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-10 08:41:51.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/05/869305/2/check/iscsi-driver-alletra-6k/80fd298/80fd298.html : FAILURE in 1h 27m 24s","accounts_in_message":[],"_revision_number":2},{"id":"44a429fc2a67a6992520ce7d3cf752357675182f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-10 08:45:24.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/869305 : SUCCESS in 1h 37m 10s","accounts_in_message":[],"_revision_number":2},{"id":"f39d9b4477eb982549e54770d4535034166d4557","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-10 08:48:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/869305 : FAILURE in 1h 47m 29s","accounts_in_message":[],"_revision_number":2},{"id":"18ac4018432d573012c5abcb476147d26b73284f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-10 09:00:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/869305/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/e2de9ca/ : SUCCESS in 1h 38m 12s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/05/869305/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/db5f8ca/ : SUCCESS in 1h 55m 09s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/05/869305/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/154ab38/ : FAILURE in 2h 08m 02s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/05/869305/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/2495747/ : SUCCESS in 1h 47m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ceae487d23ccfad41b39c12991f8be80f3100b8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 09:09:46.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/db7777e320b645119663223ab7e08bf5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dd9a0fc1a58419ba94f615e811a4bf8 : FAILURE in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa74f54c0976493b978c6454814c9358 : SUCCESS in 8m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/915bf5b88aea42439262e3745b8ee88a : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51431925c6ff4a83926a6ffa3e951282 : SUCCESS in 14m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/693fa36754944d6995d5de24ba97a2d9 : SUCCESS in 40m 16s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2a02937fbb2040639c56b9bd20f9be07 : SUCCESS in 49m 18s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c939213cba4b432d98abb6bc983d6ed3 : SUCCESS in 1h 30m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56487aea62d3418299222061de81649f : SUCCESS in 1h 02m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcbdbd02f69a474fb477c76a00a1787c : SUCCESS in 6m 41s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4f9bb11fc1734f42a4ce4e018f6503ab : SUCCESS in 12m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f25e5a880f7b4c859f726449a8ac0656 : SUCCESS in 6m 05s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0134bd4a7b2949b4aef27fb842e70377 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/519eba640e0c452d9eb015f249caa9b9 : SUCCESS in 5m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1db8ce43533249ad8a7da6e3e3246253 : SUCCESS in 5m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/646d6c31b2734c60a4b6ea4fef9fb7e3 : SUCCESS in 34m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e5498a2aa48c49769d59c70fee5234e5 : FAILURE in 4m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4cfd3b47eb194e12bde8deeb4c1b86dc : TIMED_OUT in 2h 04m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/188791aa00034d1cb5fa7bcac1404009 : FAILURE in 1h 17m 24s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8be16be9acf64e7db809115d7a90ac0d : SUCCESS in 2h 13m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d075baf8b3024d1682a4eb2ccb739b50 : SUCCESS in 50m 12s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/77e697b5f3e74c968c40ad5b6c32d7d6 : SUCCESS in 1h 10m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/99c4fc6d7b064f109939fdf7f7159b53 : SUCCESS in 44m 01s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1a05906a67f64cc0906357784ab07eb9 : SUCCESS in 1h 49m 59s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1c8297614c7f40b880bad8766eab8ff4 : SUCCESS in 1h 54m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/869811c0a5294646a404e270846e2a0b : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":2},{"id":"13f8dcf51c77b956c1b0326bff51b930e6f4e7e6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-10 09:21:32.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/05/869305/2/check/DellEMC_Unity_FC/9e242ff/DellEMC_Unity_FC/2121 : SUCCESS in 1h 28m 23s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/05/869305/2/check/DellEMC_Unity_iSCSI/fb6a5cb/DellEMC_Unity_iSCSI/1992 : SUCCESS in 2h 30m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"adbd07426d29aec34d0d804b00416c6f82c01252","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-10 10:06:45.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-05-869305-1/ns5_nfs : SUCCESS in 3h 19m 17s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-05-869305-1/ns5_iscsi : FAILURE in 3h 27m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"7d816091f7c24146a3638e84ebf289c0272fb4d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"date":"2023-01-10 10:23:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d0786050d28ce54f741dd40f2c1bbf431854006b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-10 10:24:16.000000000","message":"Patch Set 3:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7bde67954b29f5cbe9a37d895fe22c2ac71dacc5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-10 10:28:05.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-05-869305-2/ns5_nfs : SUCCESS in 3h 17m 38s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-05-869305-2/ns5_iscsi : SUCCESS in 3h 25m 46s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"2dbc535c124fdf8f7f66c50ea09ad5f96ee6f2ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"date":"2023-01-10 10:50:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c9b667532a1a7c2988e2c4f8ec8340df839ec8d5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-10 10:51:11.000000000","message":"Patch Set 4:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5d74673ef2d0d41812434145215472db7660c8d3","author":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"date":"2023-01-10 10:54:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5c4444c01ad6a4dd62c5fa7265e864c55d4f4c08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 12:01:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e05e6f9473440869da391aa49c8f643\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f6ed4c098694323994f2af35d22fcdf : SUCCESS in 27m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7528436f73484e569162eef1b2068910 : SUCCESS in 20m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"81c0ee240a618848183c9d319262c246049f38a3","author":{"_account_id":32465,"name":"IBM Storage CI","email":"hchintha@in.ibm.com","username":"ibm-svf-user"},"date":"2023-01-10 12:25:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- storwize-tempest-dsvm-full-iscsi https://ansiblecinder-logs.s3.amazonaws.com/storwize-tempest-dsvm-full-iscsi/8ef0632/index.html : SUCCESS in 1h 30m 58s\n- storwize-tempest-dsvm-full-fc https://ansiblecinder-logs.s3.amazonaws.com/storwize-tempest-dsvm-full-fc/3e3e9a3/index.html : SUCCESS in 48m 29s","accounts_in_message":[],"_revision_number":4},{"id":"21aa6ea811460328bd93403f1e0c1d34a0a5d517","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-01-10 12:29:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/869305/1/20230110153917/NEC-V-ISCSI : SUCCESS in 2h 46m 44s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/869305/1/20230110153917/NEC-V-FC : SUCCESS in 5h 49m 43s","accounts_in_message":[],"_revision_number":1},{"id":"f0d38ebe6038ad2173dd556ee352f964fb037a3d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-10 12:40:21.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/05/869305/4/check/iscsi-driver-nimble/5ef3ceb/5ef3ceb.html : FAILURE in 1h 21m 43s","accounts_in_message":[],"_revision_number":4},{"id":"7e74259218faebf1aa296e384a33fc190defba9f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-10 12:47:50.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/869305 : SUCCESS in 1h 27m 04s","accounts_in_message":[],"_revision_number":4},{"id":"d667ddfbb914f84ec56c2afe958251f425e4470f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-10 12:50:31.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/05/869305/4/check/iscsi-driver-alletra-6k/7b00cb4/7b00cb4.html : FAILURE in 1h 45m 29s","accounts_in_message":[],"_revision_number":4},{"id":"ed5e1e084f698d8b00bc23dd2765be6a666ace0b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-10 12:52:06.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/869305 : FAILURE in 1h 30m 52s","accounts_in_message":[],"_revision_number":4},{"id":"a3085031a5f3f37dfcbf4222b1f0fec213fe5295","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 13:05:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e54fd02bc72a45b98b0c30e00cb7f09d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af20dfb2156846529bc774fb2fa4bd56 : FAILURE in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f17118a646c84da4a87b572ade1ae850 : SUCCESS in 10m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a4b9b49a0e14148861c3d63bfd2e49f : SUCCESS in 10m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84e191d72abd442487a47d45c5ffb072 : SUCCESS in 12m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/db6b7f64122a4e33b1a72d1fb9c53f9c : SUCCESS in 59m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ebcc5563a0844701ad63eeec9e6ebf37 : SUCCESS in 59m 08s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c002c03d35284abb962f7959ee5ba3b8 : SUCCESS in 1h 39m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/987289852cc4497ab3c40ef018ede8f5 : SUCCESS in 59m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6d9871c7575415bafa1fe2143c621ca : SUCCESS in 8m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ba7c9596a98440d9ac7d5252d9be579f : SUCCESS in 15m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/56542e72b3714cc291cd94ca70654f68 : SUCCESS in 6m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/01482ccd73f64bae8924d9bfade4b795 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8b2bf8713d304052a9b950bdcf4aaa11 : SUCCESS in 6m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/575e93c705594bd0b0931beffd481c9c : SUCCESS in 7m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/acd897c7d25d42d88274ff4f6248b00f : SUCCESS in 48m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0190335f3fcf41d7920f90e9d35eed4c : FAILURE in 4m 24s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/adc1bcd365f44dce998ce9a977ae76bd : SUCCESS in 59m 03s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/32fdb63d78594179a81df9d3bec59b4c : FAILURE in 1h 38m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/15c4afee02944269aa025df5a3646347 : SUCCESS in 2h 03m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/115f81877502430ab42f0eb17a0a6e58 : SUCCESS in 59m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/03ffb1ef64584302af8be3af2d851221 : SUCCESS in 1h 38m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/908b7eda03e04083b5c049fe8616ac5e : SUCCESS in 40m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4161b30256ed484ba7c254b61aef571d : SUCCESS in 1h 44m 25s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6939b5de34a440a59de1db968765397f : SUCCESS in 2h 09m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1de088d287b46848f2c4bde36da51fa : SUCCESS in 1h 03m 19s","accounts_in_message":[],"_revision_number":4},{"id":"987c9c3ac2d77fc14b5aa51edd39baef774ad54b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-10 13:36:02.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/869305/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/11015f0/ : SUCCESS in 1h 39m 53s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/05/869305/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/b43581f/ : FAILURE in 2h 20m 23s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/05/869305/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/32bb58c/ : SUCCESS in 1h 39m 54s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/05/869305/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/d5cd883/ : FAILURE in 2h 17m 20s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy pure-devstack-cinder-tempest-nvme-roce-aio-jammy : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"781885c52652fd02804200cec50324fe7dcca9f3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-10 13:55:29.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-05-869305-3/ns5_nfs : SUCCESS in 3h 27m 56s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-05-869305-3/ns5_iscsi : FAILURE in 3h 31m 29s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"89f825c6df6d64bef362ae6013c66e5e255df656","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-10 14:21:25.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-05-869305-4/ns5_nfs : SUCCESS in 3h 28m 0s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-05-869305-4/ns5_iscsi : SUCCESS in 3h 30m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"35cb20cdbf5a5983af202342bd6fa6bcccffd6e0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-10 17:15:01.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/05/869305/4/check/DellEMC_Unity_FC/84f53c2/DellEMC_Unity_FC/2124 : SUCCESS in 1h 40m 13s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/05/869305/4/check/DellEMC_Unity_iSCSI/ce7070e/DellEMC_Unity_iSCSI/1995 : FAILURE in 2h 33m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"c912b63b28bff0c488938692e42ffcc39459612b","author":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"date":"2023-01-11 05:46:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a2823005656aa28e07b9b356bbcea7f23245a9be","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-11 05:47:29.000000000","message":"Patch Set 4:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c5f338287056d1b16eec6933a23858c417390e41","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 06:12:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86fca1f42998431c9082026b6c5b6779\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/26c0c1061d9a41a5bcfed27bba10c592 : SUCCESS in 23m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c4958ba22e4e491fb44a42e243c4b4f2 : SUCCESS in 21m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e3aa9c7cbe93e165ccbf4413b457a41d0ba14805","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-01-11 07:53:48.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/869305/4/20230111100748/NEC-V-FC : SUCCESS in 6h 03m 58s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/869305/4/20230111100748/NEC-V-ISCSI : SUCCESS in 6h 45m 57s","accounts_in_message":[],"_revision_number":4},{"id":"6d09f3405990e26f5667ad3ad3bc184364470ae5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-11 07:55:20.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/869305/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/e48a78a/ : SUCCESS in 1h 39m 23s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/869305/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/84b42da/ : FAILURE in 1h 57m 19s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/05/869305/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/ebbdb3f/ : FAILURE in 2h 07m 47s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/05/869305/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/e6dfff7/ : SUCCESS in 1h 42m 28s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy pure-devstack-cinder-tempest-fc-aio-jammy : NODE_FAILURE in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy pure-devstack-cinder-tempest-nvme-roce-aio-jammy : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"baf7b82a6b24357401a3ba840c11ff5618a74c4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 07:59:44.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bc6afae93e0460384c597c8683321ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f2e3d13b81a4321a335d6d7ba651be4 : FAILURE in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/130ac381b9f54f4c8492f7d744a37b55 : SUCCESS in 11m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/625c21ea8b8d4404a5b8f2af45c00e26 : SUCCESS in 9m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c7db392ac4941ddb8c4929a5047d88a : SUCCESS in 14m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/3012be8f73bb4ac5be07bb14ad8fcf9c : SUCCESS in 49m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/711a44383fbf4a9397096d0db4fcd8c4 : SUCCESS in 53m 50s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1d3a03256b6d4825af64cf3868d701c9 : SUCCESS in 1h 11m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19921601fd9c467bb918144fa17f9a06 : SUCCESS in 50m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02fe88de24554babb3b8d75355356002 : SUCCESS in 8m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fc546adb786a4cee88530c993edee4e3 : SUCCESS in 18m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/69cddfc6224c41acbf911f3967de4cfe : SUCCESS in 5m 33s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/12af9b65489f4ef4a25edb151ac8639e : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a6da46854abc4cf69ca97cc5821261c1 : SUCCESS in 5m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4ad8b753f00e4d3b915700be76479270 : SUCCESS in 5m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/59d8e69fd9744f4caa9bea8bab72d403 : FAILURE in 15m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fe8d491a825d4c6f8c5018c88db57fb3 : FAILURE in 4m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/43d582a3881b4e8a888938cf8005d843 : SUCCESS in 1h 16m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8d1c0331b80c4d73a434a7d2de27a70f : FAILURE in 1h 24m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/548f7a6d8bb6492b986f0f625f81696d : SUCCESS in 2h 09m 28s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/470867909b934d308c9ebfb01af0f256 : SUCCESS in 1h 16m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6a79bdb5c7784998a7c38af8d4804bab : SUCCESS in 1h 48m 40s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8017a5a5a08842cb8006035d8406e4e0 : SUCCESS in 38m 51s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d44225cc3e2f47a7b1bd11fcb027650d : SUCCESS in 1h 33m 27s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3338f60555af4e2fb8cf62d44aa6d481 : SUCCESS in 2h 12m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4564e54b87e94edd91b7554da4cff3cc : SUCCESS in 54m 41s","accounts_in_message":[],"_revision_number":4},{"id":"37c5264a50a7acf54e985c89ba608bb156c4f053","author":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"date":"2023-01-11 11:57:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1375f701287e5f6df84acaf012947d64fd4a3d1a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-11 11:58:38.000000000","message":"Patch Set 4:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8f06fd402af84850235f4da95d91e9a6538b65f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 13:55:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85156251c16544e485ae5368b94b6725\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0267e94bb0ae43d9991ca078e7aa4173 : FAILURE in 29m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7406822e9ee74d628dc1a72c97946a7e : FAILURE in 35m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"adb08466dc73f824b9e2b3231a0b2a53ca3d5b68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 14:07:19.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/edb67906e7694faea8ea47fbc17d8d0b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bc82f73398040dd9692c52d7437b77c : FAILURE in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efa6d51eac9741f0b8595362d43e03f7 : SUCCESS in 11m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f8d9a23119541288db5233314792b4e : SUCCESS in 11m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e239a647581d4deca3f52086f2a1a888 : SUCCESS in 12m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/e43cf66ccaf4484bad7f7aabb6404afa : SUCCESS in 48m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/04277a0ce3e84c1c90e61775c2ad431f : SUCCESS in 54m 45s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ebdba7b6da8f47bba8932f218257669c : SUCCESS in 1h 12m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37060c56ad814e8998b7f5e7c56abbde : SUCCESS in 51m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96f5f6c6d6b5427988e9dad401d40efd : SUCCESS in 8m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8f11190ac2934d4d8fdb198c75cdd3f7 : SUCCESS in 15m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c9cffd043bc740efa9178db24470c20b : SUCCESS in 4m 36s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/83d31c7138bd4a549d1875b41678ee03 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0c2bff6d542240c195df37e4f37e0f4f : SUCCESS in 6m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f7de357c427445c3a7d6ca7b2aea26ca : SUCCESS in 6m 25s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9159e24e995d44bd8d9a69d61d609f40 : SUCCESS in 35m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d26f5d9f9c4049788975329199c6f31a : FAILURE in 4m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e7c80df0e66848d78f2b39a814063c72 : SUCCESS in 1h 30m 04s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a907a3ecd3994898a189eb2e8d06ec29 : FAILURE in 1h 19m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b1efcef9d62a4b5898d599a54ae96bf2 : SUCCESS in 1h 37m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2cc5af7fa78e4b98a3f1202ab9bb5cbf : SUCCESS in 1h 08m 27s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ebce5584c98243c39158713b99d3bbab : SUCCESS in 1h 25m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bdf1c9d4a1b14c7ea86f8ee5cf71de1e : SUCCESS in 39m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f3f5d09cf4204e919bfa29774c8023d1 : SUCCESS in 1h 21m 49s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b33a35cca507487cb5e4f6f770bae56e : SUCCESS in 2h 04m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0566ba393f444b7a836028667a9d0245 : SUCCESS in 48m 38s","accounts_in_message":[],"_revision_number":4},{"id":"b572422a81b7afe6f411e03d4b8e50328bba4e2f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-11 14:17:53.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/869305/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/13d48e4/ : SUCCESS in 1h 36m 53s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/05/869305/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/9852075/ : SUCCESS in 1h 45m 32s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/05/869305/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/4570a7e/ : SUCCESS in 1h 46m 11s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/05/869305/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/0d72d2c/ : FAILURE in 2h 19m 12s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy pure-devstack-cinder-tempest-nvme-roce-aio-jammy : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6a6f7cee5ee849b18533a391bf1d3c3bb3b19e2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"date":"2023-01-12 07:18:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0a388806892bbe054c06b1245a43d35fed25a2f4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-12 07:19:01.000000000","message":"Patch Set 5:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a7c95f9044caea3989efa8226a787f98876510ae","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-12 07:19:27.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"b31ae58cc57cfd4c5eb00e0739ff611f13ccb740","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-12 07:39:16.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"3e36226aa6577b7eafcf4ed24aa1355911604d7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 07:48:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/285cfed33a5a4f5bab4b595a685df1ba\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a9b5b78ba0347c7829d35ba3ded674f : FAILURE in 28m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/07cedbbb2ca74fb4a8b8eef8c141c8c4 : SUCCESS in 19m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e2c0c2253a23521c9cd0b9d2b844edbe9ec56c31","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-12 08:18:15.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/05/869305/5/ : FAILURE in 51m 42s","accounts_in_message":[],"_revision_number":5},{"id":"416d0d6ca92f869370b9f4a667812fd02d424811","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-12 09:10:00.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/05/869305/5/check/iscsi-driver-alletra-6k/6a98af0/6a98af0.html : FAILURE in 1h 41m 05s","accounts_in_message":[],"_revision_number":5},{"id":"6d6dee0591f2bb707a3f5cb0a892387f821f83d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 09:18:12.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4268410d4364690934c9e26a814591b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aa781b2bc8843f1a1cd7231d90c425c : FAILURE in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a73d70137904946b1fefe961e682a8b : SUCCESS in 11m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8facf0d3a0f54f49aa429274ed46bc76 : SUCCESS in 10m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/774a357afcd1437790b34c3f6e04a0f9 : SUCCESS in 14m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/ac6c3cb8a8d340ddb724df3e9dd8b862 : SUCCESS in 54m 25s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/aa8983fcce4d430ca9f927e0478a8020 : SUCCESS in 1h 01m 31s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b12bb5044d8845a7abd03e35f6b898ad : FAILURE in 1h 15m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ccb8e1c79ec043e4872cfdc26b601b39 : SUCCESS in 51m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85df25b4c98c42c289e5f047b917b1b4 : SUCCESS in 9m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/095c7da09a6249dd897f070c21f40a51 : SUCCESS in 14m 42s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0494e339da6f44a0b9132baf6c7100f1 : SUCCESS in 5m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/183b0dc9c24f49f9a0adadeb06a0260f : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/317f95195a9d4b31a661fe3804159fe8 : SUCCESS in 7m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/addf5731513b4c03a3188380565c3adc : SUCCESS in 6m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8b21f11e63d94f0393f8c983d4d2954b : SUCCESS in 44m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/83876c9c866144149d2fe5d8a99646c7 : FAILURE in 5m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4b9038ca3c4044bd94bd505bbef90689 : SUCCESS in 1h 23m 28s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f3806a94cb0a45b7946b95c8b5fd8976 : FAILURE in 1h 12m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7db1022fbdd9479d91084ca5b2973adc : SUCCESS in 1h 52m 32s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7038eec9baf74be9897681f37393dc34 : SUCCESS in 1h 17m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dcf12e04abcc4576897b5419cb837058 : SUCCESS in 1h 32m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ce68aa89670f43549f904af3a35fb499 : SUCCESS in 44m 11s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c1c454e54ef44829a980ab681b9824e4 : SUCCESS in 1h 30m 22s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5ef7a0656cc9436a8bc1c0c9016393dd : SUCCESS in 1h 57m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8face56397a04ffdae6b338b816f74ff : SUCCESS in 59m 15s","accounts_in_message":[],"_revision_number":5},{"id":"2d63678be2291bc35b599e2fc0f9c1da2ec70bb6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-12 09:35:09.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/869305 : FAILURE in 1h 29m 01s","accounts_in_message":[],"_revision_number":5},{"id":"65ec827fd6b551cdce55c68212dda88557ba49bd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-12 09:43:44.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/05/869305/5/check/iscsi-driver-nimble/f48f67e/f48f67e.html : FAILURE in 1h 21m 17s","accounts_in_message":[],"_revision_number":5},{"id":"5388ba62264978d17909c48953775f1ac9049bad","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-12 09:45:29.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/869305 : SUCCESS in 1h 58m 28s","accounts_in_message":[],"_revision_number":5},{"id":"5051b24b0f53b7fe24f1549d9edc86ac9f18e79c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-01-12 09:56:57.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/869305/4/20230112120804/NEC-FC : SUCCESS in 3h 35m 58s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/869305/4/20230112120804/NEC-ISCSI : SUCCESS in 6h 48m 49s","accounts_in_message":[],"_revision_number":4},{"id":"3a677d2d6e49683f67e736b986e6ac6e7de97291","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-12 10:13:05.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/869305/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/bc5a079/ : SUCCESS in 1h 38m 34s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/869305/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/9aee168/ : FAILURE in 2h 53m 50s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/05/869305/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/ff2b189/ : FAILURE in 1h 52m 55s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/05/869305/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/deb798d/ : FAILURE in 1h 45m 06s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy pure-devstack-cinder-tempest-fc-aio-jammy : NODE_FAILURE in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy pure-devstack-cinder-tempest-nvme-roce-aio-jammy : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bb75841548a48cd88de6bac56a661274e532d4aa","author":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"date":"2023-01-12 10:38:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f64b915b0af01bcb06803eef423f3b04e0f4dc4c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-12 10:39:07.000000000","message":"Patch Set 5:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f60c2dd82c15603c792e69060f4549521d05a22e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-12 10:49:18.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-05-869305-5/ns5_nfs : FAILURE in 3h 30m 34s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-05-869305-5/ns5_iscsi : SUCCESS in 3h 28m 21s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"f1a6325885acfe88edca1d3f12f9eb307bab1908","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 11:17:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48e50b223a5b418b9db766407a89838f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e0612a91cbc481db0cbb6629ad760cb : SUCCESS in 24m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e620280d6e9644c98c6accc38b2761bc : FAILURE in 26m 13s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3c1b5b851958417c08abac988f6fd2bb7c139540","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-12 11:54:45.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"347566eca838788d6f4e87277479416816f4b7ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 12:47:41.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c08610ab85474a8ba5b3f7561741f957\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a25de9ef548a460bba4263d28fd8c6eb : FAILURE in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21f16591652b4eec8987d2de18e6e53d : SUCCESS in 13m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b987ced9ffab4e749a825eb2b5e1d941 : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9c5bf2644cc439596fc5b02be074cfa : SUCCESS in 14m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/75b83204b06144be92d1dea5c95b8dff : SUCCESS in 1h 11m 35s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/60800dd648c34a2a8eb8ead7dc17c8af : SUCCESS in 1h 00m 08s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/23abf537d09349f9a071da1b40bfb1ea : SUCCESS in 1h 34m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a119b51410b4eacb7b23c73a05d41fe : SUCCESS in 1h 01m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3ef6c4a89a545e2ad9da06886968dd5 : SUCCESS in 7m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9aad6d3a05e54d1ebd979a42eae6324e : SUCCESS in 15m 01s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b539de82a6244c37885e6e39d99035ef : SUCCESS in 6m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/528fc28b82184a4a993d95eb1557bd6a : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4da759b7c6214412bb51bdd19d442d85 : SUCCESS in 6m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0bb000cc3bb54c51aeb28e3d5f52f9b3 : SUCCESS in 7m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0ecf2bd047554f0b8d43025b640b1982 : SUCCESS in 52m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2002ad3ba3f5416584c905159c0a81d0 : FAILURE in 4m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/15333ce4a7a34194a0c324fca1bcf0d6 : SUCCESS in 1h 09m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/51a969f54b664349a26000609b750c11 : FAILURE in 1h 29m 17s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c4585b4ec05345d3b77864fda1939d8c : SUCCESS in 1h 50m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9678828257144effa2fab56455586f23 : SUCCESS in 1h 10m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d8288bc4f6064e62bd8c840fd8d88104 : SUCCESS in 1h 24m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6bd6ec516a62492fbd5fbea86b5ab7df : SUCCESS in 39m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b030912ede9644e29cdb51e6a0f15d8b : SUCCESS in 1h 29m 30s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ee324be32e384129a7676c34817b3828 : SUCCESS in 2h 03m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9908b27de3274d61ae76ce61c3efc4b9 : SUCCESS in 46m 48s","accounts_in_message":[],"_revision_number":5},{"id":"bfdad62aedde4cb65a42cf9daf768b2f2ea5b2c1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-12 13:36:15.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/05/869305/5/check/DellEMC_Unity_FC/b4253d0/DellEMC_Unity_FC/2139 : SUCCESS in 1h 39m 07s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/05/869305/5/check/DellEMC_Unity_iSCSI/1a0550e/DellEMC_Unity_iSCSI/2009 : SUCCESS in 2h 52m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"b3db3fd0cb6a9894d6cbfbc74c2b18689e2bb0f2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-12 20:01:20.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/869305/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/8a240d6/ : SUCCESS in 1h 44m 29s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/869305/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/57879ce/ : FAILURE in 1h 54m 00s\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/05/869305/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/44f2794/ : SUCCESS in 1h 52m 17s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy pure-devstack-cinder-tempest-fc-aio-jammy : NODE_FAILURE in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/05/869305/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/e653ff9/ : SUCCESS in 2h 03m 13s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"420468e5c927683084f6fabfa92b8f8a28de20b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"date":"2023-01-13 05:22:37.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c3d2436484210f93a3911f62508de76ffa9d918a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-13 05:23:23.000000000","message":"Patch Set 6:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e78152a54fd3a873f034aa3e4bcf31115390be96","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-13 05:24:32.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-05-869305-6/ns5_nfs : FAILURE in 1m 24s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-05-869305-6/ns5_iscsi : FAILURE in 1m 21s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"3a8806a1af94adfb1b29711a54769f6aad35ca06","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-13 05:39:05.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"7c247f2e13f5fdf601bd2a2d6b3f23905ea3e478","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 06:01:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8c77a8190d143639d3d48afa34e8ac1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/484c458538a1449f9d7b64b97a65c330 : FAILURE in 34m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9d47d1f9e124451eb04d3cbf223f2c53 : SUCCESS in 35m 50s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a790efe7cfc252d26ae72a2b72ab3ad5fb7e9bd7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-13 06:13:43.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/05/869305/6/check/iscsi-driver-alletra-6k/69ef48c/69ef48c.html : FAILURE in 25m 32s","accounts_in_message":[],"_revision_number":6},{"id":"08c8aade53e234c6576447b5f465924c5f012eaa","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-13 07:23:56.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/869305 : SUCCESS in 1h 48m 55s","accounts_in_message":[],"_revision_number":6},{"id":"f570317f9f06b53656368055398d031ec79aae13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 07:52:52.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/35b7d9904fbc47159e2d4bff280cf934\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1f1e1966f1b4cbcac2f8226dbde562a : FAILURE in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e8905dc8fb94bc886ad7f982b1e1cf9 : SUCCESS in 9m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/310047e9a02f4591a5e0b48ff667b899 : SUCCESS in 10m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d72d27ccb69545adb5c671643820c69a : SUCCESS in 12m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/e96b29083d694de6b015f0c5d3a96f07 : SUCCESS in 52m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fe307c0ea38149229b21469cd76ff8da : SUCCESS in 54m 07s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a2b207f189b940b9b7f5c3cbaf24c534 : SUCCESS in 1h 05m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01efef4caefa46d189882d6d0221d662 : SUCCESS in 57m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6262070be2504ff090d0bfffaf704192 : SUCCESS in 7m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9a1938354a654d01b295f0ec63430fa3 : SUCCESS in 12m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ce3b6c2026ae4217b615fdde17360bc7 : SUCCESS in 5m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3db4ec44b8a143f8b50a776cf033398a : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cf440c49c87d43f2bf6270e78cdc454b : SUCCESS in 6m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7c413c12bb864dadab011290ed39a379 : SUCCESS in 5m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/647ffc76fdd44c2a939125d80759c484 : SUCCESS in 38m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/58642a497c9646b08c61991764274cd3 : FAILURE in 4m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/93a69beb45014a38b7724a4eb91a56ae : SUCCESS in 1h 06m 00s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/24e54b7afa5b43b8be3fb6d1ad8be30d : FAILURE in 1h 36m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ca196beaeb3d4ed798405df81392971e : SUCCESS in 1h 25m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/182e6c635faa4893b6f8eb1a5172cccc : FAILURE in 44m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5711e61713144a6ab5ecde8f9ae544fb : SUCCESS in 1h 44m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/69e3e242de4c48c4bc06fac3120d9d1a : SUCCESS in 32m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/125096081484424a82cf5b9e16420c1f : SUCCESS in 1h 42m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/05c9512314c6449a9a94dd09fee6fe74 : SUCCESS in 2h 14m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a71c0d3b1e44da1b88d32666a113897 : SUCCESS in 58m 53s","accounts_in_message":[],"_revision_number":6},{"id":"c98658ddda671d6c499b190452e17caad902a436","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-13 07:55:30.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/869305/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2f1060b/ : SUCCESS in 2h 30m 17s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/869305/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/db3de28/ : FAILURE in 42m 31s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/05/869305/6/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/a6f25a4/ : FAILURE in 1h 09m 09s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/05/869305/6/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/107e48c/ : FAILURE in 1h 08m 36s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/05/869305/6/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/f4dc575/ : FAILURE in 50m 04s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy pure-devstack-cinder-tempest-nvme-roce-aio-jammy : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7566aed25252d2f0ed0098e4c7b850fe97ff369a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-13 08:18:36.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/869305 : SUCCESS in 1h 30m 36s","accounts_in_message":[],"_revision_number":6},{"id":"1fa4f566e506599759940adb8e6592d679b67b9f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-13 10:09:01.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/05/869305/6/check/DellEMC_Unity_FC/ee22cbc/DellEMC_Unity_FC/2148 : SUCCESS in 1h 35m 37s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/05/869305/6/check/DellEMC_Unity_iSCSI/f1aa154/DellEMC_Unity_iSCSI/2017 : FAILURE in 2h 55m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"fb033dc55e3d487a6ce9521139c7aab578dcc9e4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-13 15:51:05.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/05/869305/6/check/iscsi-driver-nimble/a0bccd2/a0bccd2.html : FAILURE in 1h 46m 46s","accounts_in_message":[],"_revision_number":6},{"id":"519908c364e3eed0be8413828fb930c5f0533b4a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-01-15 03:04:14.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/869305/6/20230115054927/NEC-V-ISCSI : SUCCESS in 2h 44m 27s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/869305/6/20230115054927/NEC-V-FC : SUCCESS in 6h 14m 43s","accounts_in_message":[],"_revision_number":6},{"id":"0bddcccc1c64ec2bc8d78ccbc3a479afb33ed054","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-01-15 09:41:14.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/869305/6/20230115150601/NEC-FC : SUCCESS in 3h 33m 26s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/869305/6/20230115150601/NEC-ISCSI : SUCCESS in 3h 35m 09s","accounts_in_message":[],"_revision_number":6},{"id":"6b4412c4b2451f614f2407c505c8cb6b86753a98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"date":"2023-01-16 08:06:24.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"44db61b008ecc2dbef73f9a03cef1706f3ab4c9d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-16 08:07:10.000000000","message":"Patch Set 7:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":7},{"id":"90fb422d0e7bf5414a5f54df0608f7f31231e78b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-16 08:21:52.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/05/869305/7/check/iscsi-driver-nimble/d2bfedb/d2bfedb.html : FAILURE in 14m 25s","accounts_in_message":[],"_revision_number":7},{"id":"ce74f2419f5bf9c0293d54edf8aa5a145cfae6ad","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-16 08:21:56.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"7502d27a23ce4314051560af56c0353d6e248841","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-16 08:35:07.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-05-869305-7/ns5_nfs : FAILURE in 28m 12s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-05-869305-7/ns5_iscsi : FAILURE in 27m 46s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"4859e07026cf26df187598e9d6363381f1805de2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 08:35:08.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa1e15aa38a94d86bfe5b0f9ddb5de04\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9aeb3f32647c413cb539565430bb69b2 : SUCCESS in 26m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/96fddc87184f464fb82aa8b3651bcfc9 : SUCCESS in 25m 28s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4dcb8ffddcce70eded98f0e03ac89d1a75a29490","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-16 08:44:32.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/869305 : FAILURE in 25m 56s","accounts_in_message":[],"_revision_number":7},{"id":"bc1a56f3d06a416f5ff0db8e27443a1c643afa7f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-16 08:49:03.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/05/869305/7/check/iscsi-driver-alletra-6k/df12c3c/df12c3c.html : FAILURE in 26m 07s","accounts_in_message":[],"_revision_number":7},{"id":"db3bea8c16cd4782521e70c053f330011606c318","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-16 08:51:54.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/869305 : FAILURE in 23m 22s","accounts_in_message":[],"_revision_number":7},{"id":"54a0f34293db96ae77e212608e86d5cf39fad71f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-16 09:58:25.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/37ad378433a845769c242f56a8949314 : POST_FAILURE in 1h 41m 33s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/e9058acc4dd7444186e6bce0496a7f75 : POST_FAILURE in 1h 32m 38s","accounts_in_message":[],"_revision_number":7},{"id":"273e477d1806d1dc3ea8eccf975d5b3d9d8339f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 10:29:48.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0a1e5eb13f04642a50386c1f626d56d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f75a6b47f5e482780b40b7bb1bdc4af : FAILURE in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93374c6df3df4b709c86069b703f255a : SUCCESS in 10m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a03dd6247b74b4f8f6e6e0882e78e9e : SUCCESS in 10m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a1f5048ec474c0899394242b6adf26d : SUCCESS in 11m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/ba379094c3004da5ac1f5d09aa6d47ef : SUCCESS in 43m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6f27f8a9e5d5404295effa3682b3066e : SUCCESS in 1h 13m 35s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fb43b42456414cca9b5540869a41ac3f : SUCCESS in 1h 35m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66136d93734842798a8b14701dde0188 : SUCCESS in 58m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb17947b57304bd08039b75dcc8c39c4 : SUCCESS in 8m 36s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2e0f284e78064c07acfa0f3703c68b4c : SUCCESS in 17m 38s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/52e3cd47065745a48bf6e29076f55e2f : SUCCESS in 4m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/120190a2da3f4aeeb0b1663cddcf2a39 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6fa2e22273e14916869ceb8f22911b9e : SUCCESS in 5m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f428b87049684ce99f2f9d42e3421c2c : SUCCESS in 6m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/c31e6ca86d29406485a89db33ca65e35 : SUCCESS in 46m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f819258ae1c34c489e7d01a18a4f60fc : FAILURE in 4m 28s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4ebe775cd4d047e08917022240cf80fe : SUCCESS in 1h 33m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7e04acd9d1794669982a0f2cbd2a2dae : FAILURE in 1h 21m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3e458545660f440caa5b522ffbbf072b : SUCCESS in 1h 48m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/06050dda3c9e4fafb4c44f14c870da39 : SUCCESS in 1h 20m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/60b8153cf56c4547840ce8a99747f51b : SUCCESS in 1h 25m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/ccf1b762b440427ab18f8bbfe916e2c0 : SUCCESS in 39m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b307e94a285545a59f841bd6150d7a62 : SUCCESS in 1h 43m 53s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/60e004bc1ff34cff85d4058823cb8b6a : SUCCESS in 2h 00m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fea83335a5fe4464a44318310e0b85fc : SUCCESS in 47m 17s","accounts_in_message":[],"_revision_number":7},{"id":"3bc3ac03aa03144ce78be90ec8457af783ce0bea","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-16 10:41:53.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/869305/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/4848106/ : SUCCESS in 1h 57m 47s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/869305/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/fd6d078/ : FAILURE in 2h 33m 25s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/05/869305/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/cb1af9b/ : FAILURE in 1h 57m 20s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/05/869305/7/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/3e2fb13/ : SUCCESS in 1h 49m 19s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/05/869305/7/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/9d54a46/ : FAILURE in 2h 00m 16s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/05/869305/7/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/9ac9aa0/ : SUCCESS in 2h 00m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"16ba87652cc2b853bed5fb3e818eb0a0bc48b09f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-16 11:05:14.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/05/869305/7/check/DellEMC_Unity_FC/fe06123/DellEMC_Unity_FC/2153 : SUCCESS in 1h 41m 43s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/05/869305/7/check/DellEMC_Unity_iSCSI/b54f300/DellEMC_Unity_iSCSI/2022 : FAILURE in 2h 58m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"d913a6134e1bb6a2fdfd5285509029216050dc5c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-01-16 17:42:34.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/869305/7/20230116210335/NEC-V-ISCSI : SUCCESS in 2h 28m 03s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/869305/7/20230116210335/NEC-V-FC : SUCCESS in 5h 38m 55s","accounts_in_message":[],"_revision_number":7},{"id":"683e082df22071224b701c2c572386acef1d0cb1","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-17 03:53:08.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/05/869305/7/ : FAILURE in 46m 19s","accounts_in_message":[],"_revision_number":7},{"id":"90e62de5c9733a3083074a3845966579fb111dbc","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-01-19 15:13:17.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"af71840a73ec9f2c378132369c74910cf2494557","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-01-19 23:01:08.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/869305/7/20230120042522/NEC-ISCSI : SUCCESS in 3h 34m 58s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/869305/7/20230120042522/NEC-FC : SUCCESS in 3h 35m 43s","accounts_in_message":[],"_revision_number":7},{"id":"51f1a64c36cf91bb052070f3b6aefe233ac4c055","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"date":"2023-01-20 05:35:40.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"bd4d81c40c5f97516d091b5367cf11cf286f3751","tag":"autogenerated:zuul:check","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-20 05:36:16.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"373f587df6b39ba7e88cc6c7a6964ec52f61a3ef","author":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"date":"2023-01-20 05:49:26.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7cd1a03962480c0a8152a3208ecb90a773463365","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-01-20 05:49:27.000000000","message":"Patch Set 8:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":8},{"id":"c3a4390a7c571738188f4ed6e0e152551efbe2b0","tag":"autogenerated:zuul:check","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-20 05:49:51.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"5bbedb525a11292584f8d2671d85e706c71f797a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-20 05:51:30.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/05/869305/8/check/iscsi-driver-nimble/16fb658/16fb658.html : FAILURE in 14m 25s","accounts_in_message":[],"_revision_number":8},{"id":"f5b5c49c1dc156b2e941e0a2d93b4caa4b70fffb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 05:59:03.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8952104517cb4387a058595d430f4d99\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/71261a95fd1b4d06b02d9cbf39f47bb0 : SUCCESS in 20m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/03de2cdc2d724aeeb6cffc87ed712603 : SUCCESS in 19m 55s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ab9ebc5256af1412654bb1a4d371a0dadac38804","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-20 06:02:12.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a19eab35de0e469ba044a882ecf5fa66 : RETRY_LIMIT in 7m 46s","accounts_in_message":[],"_revision_number":8},{"id":"9d5b0eda3b04611f846a8cfe2925b852b54c7604","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-20 06:15:08.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/869305 : FAILURE in 26m 36s","accounts_in_message":[],"_revision_number":8},{"id":"c2e496cceb7b49007ee41209aafbf483f87b9211","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-20 06:17:48.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/869305 : FAILURE in 23m 30s","accounts_in_message":[],"_revision_number":8},{"id":"d45e6f91cc59276280c4b21d48cb0776cf0f33b9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-20 06:20:23.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/05/869305/8/check/iscsi-driver-alletra-6k/fbc82d2/fbc82d2.html : FAILURE in 25m 09s","accounts_in_message":[],"_revision_number":8},{"id":"187447655398c7e31c80ad47157d127ed6e84dea","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-20 06:29:50.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-05-869305-8/ns5_nfs : FAILURE in 51m 21s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-05-869305-8/ns5_iscsi : FAILURE in 53m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"31f74a2b12216fac0ba35d3cb9ede6fefd994f57","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-01-20 07:26:57.000000000","message":"Patch Set 8:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/05/869305/8/check/fujitsu-eternus-dx-iscsi/8ab367b : SUCCESS in 1h 50m 38s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/05/869305/8/check/fujitsu-eternus-dx-fc/f5fcd16 : SUCCESS in 1h 50m 36s","accounts_in_message":[],"_revision_number":8},{"id":"4d34e7dab2c05f52a12d8ff321f39326fb76346a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-20 07:36:52.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/869305/8/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/eaa2c45/ : SUCCESS in 1h 52m 26s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/869305/8/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/be0214b/ : FAILURE in 1h 59m 02s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/05/869305/8/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/c8bc488/ : SUCCESS in 1h 34m 08s","accounts_in_message":[],"_revision_number":8},{"id":"d4f23ad3dd3bb1d60ddebda92be1752bf6369f71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 07:38:36.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/93df54cb41dc42e19d122681449111fc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b1bab4cadc6435994dc4556935c377e : FAILURE in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cd9ef5b381945eab47c5cd271761d28 : SUCCESS in 9m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c0ae41a57cb48b7bab770fac1db959a : SUCCESS in 9m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c13fc5eae854bc895faaf6be21275b4 : SUCCESS in 11m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/a377bfe4b5fa47508c315a2fbc38e7b3 : SUCCESS in 1h 00m 38s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/55b030fc57b1417e81b388c4bd1f1e52 : SUCCESS in 43m 14s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9fe48fa5ed474276adb8eb6808432be2 : SUCCESS in 1h 05m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0537cb2773d34089874c5cfa7ea64071 : SUCCESS in 43m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/987a80b6f6d64d669d490e59a3534adc : SUCCESS in 8m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/294bd64124e34eeabb207d5d5f1963fd : SUCCESS in 14m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5aa1186c2423422194b04e0ac7c8d0cb : SUCCESS in 4m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6861ea17e01e4fe0bf385c75485e9c81 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/356699a9f7c44e318514f428e470c093 : SUCCESS in 5m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9e8dcb91621540f38eb2b7a12bc9179c : SUCCESS in 5m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b6fc292722ed4ae385590f6944a1928f : SUCCESS in 31m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/778f5cdf3869494195b7f85a3eac6811 : FAILURE in 4m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a25042c60911495b859ef4c02f036bf9 : SUCCESS in 52m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9edd2f535c07482f952ae03a825fa792 : FAILURE in 1h 14m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5bf4f32b899d41a29f7cb39ca2146661 : SUCCESS in 1h 34m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/55a155da13bd4e4299fc4caeb551edf8 : SUCCESS in 1h 03m 43s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/14b8c9e7317d4c4eba0ed22d79aa7b06 : SUCCESS in 1h 07m 37s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a3527c7e5e3c48128f584e2770b17bba : SUCCESS in 33m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1687a71455aa47dea24a2a9f41f5c186 : SUCCESS in 1h 26m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3739e6a81dfa4aceac79b796a2921443 : SUCCESS in 1h 58m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55c9c54266b84b3c86a8ec05360f5522 : SUCCESS in 49m 12s","accounts_in_message":[],"_revision_number":8},{"id":"8507050b5b7179c7b0655ff9ded297b5fe1ec8fb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-01-20 09:59:32.000000000","message":"Patch Set 8:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/969","accounts_in_message":[],"_revision_number":8},{"id":"22073e6281faa360ed8957cef8d2a0f6ae9e1c27","author":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"date":"2023-01-20 10:17:50.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"50abac1c1e40f6d0ca417f18be4e5d0317d4a1e5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-20 16:29:44.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/4a53cf28801d4889a79ed0e7adfd8806 : POST_FAILURE in 1h 47m 07s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/8d3124c1c69c489599b96d5b550597cb : RETRY_LIMIT in 1m 01s","accounts_in_message":[],"_revision_number":8},{"id":"c5aff8b7f1418b51762db7987c75367490480b7a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-22 12:39:51.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/05/869305/8/ : FAILURE in 56m 36s","accounts_in_message":[],"_revision_number":8},{"id":"ef4d91c625027764105fb14e48de6682750efc02","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-01-23 00:45:35.000000000","message":"Patch Set 8:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/869305/8/20230123034041/NEC-V-ISCSI : SUCCESS in 3h 27m 49s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/869305/8/20230123034041/NEC-V-FC : FAILURE in 6h 04m 50s","accounts_in_message":[],"_revision_number":8},{"id":"66d38960c7fe6f33e3c7ceca48a35c6e843e8a9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-01-23 15:59:46.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"1dcd3cb89fe9ce57e53ea27217ef35ef867010c2","tag":"autogenerated:zuul:check","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-23 16:00:09.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"543500111f61b2ce281b7f230490157a0dab56d4","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-01-23 16:00:49.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b25949dca68796b44fd8aba0b3c8342fcfde2fe1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-23 16:35:24.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/05/869305/9/check/iscsi-driver-alletra-6k/a18ffb0/a18ffb0.html : FAILURE in 26m 03s","accounts_in_message":[],"_revision_number":9},{"id":"d5afc8e765595322c89d5a38d2e25c8567514d2a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-23 16:42:55.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/869305 : FAILURE in 23m 26s","accounts_in_message":[],"_revision_number":9},{"id":"df896e934318e752780fdce66803246a038dbab2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 16:43:55.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6706e09d11ba4b31b285e6b741b40e1e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/18db5f632f0547ff86c35c4b8680f138 : SUCCESS in 27m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/752df9e123b947a98ca46a5dafb281d6 : SUCCESS in 22m 39s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8eb2056f9dc4e543b813c560c2ba42185786c69a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-23 16:46:07.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/869305 : FAILURE in 26m 35s","accounts_in_message":[],"_revision_number":9},{"id":"df8d88b37f4fda8c1b0aa54a85a17bba101aa10a","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-01-23 17:09:41.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/918cbe19ef81453aad6b788aee9ce420 : POST_FAILURE in 8m 05s","accounts_in_message":[],"_revision_number":9},{"id":"a709fc4af331967944efa4f554b7da537f17503e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-23 17:17:42.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7e5c2994230a43cabba267408a4d1769 : SUCCESS in 1h 16m 19s","accounts_in_message":[],"_revision_number":9},{"id":"256002b48c9f95d796b5e79e194c6736ba33897f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 17:57:51.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcf997320b21462b99626d95ab098100\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3607d25aeab84e8e9af95ad7a5cb9451 : SUCCESS in 10m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/105b34674524408eb3057620e1cc6823 : SUCCESS in 10m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52d20a0e56bc476b8e0bbb897cdad58e : SUCCESS in 9m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78ed6a4171fb4571aeced23738407ccf : SUCCESS in 13m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/166fab84ad284382ad6a6f131b861be2 : SUCCESS in 1h 24m 33s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/aa4e04db761a47bc95f6b9ca7a0b2065 : SUCCESS in 52m 38s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e3b0e3dbef6f486b8ccb364195ff10e3 : SUCCESS in 1h 43m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04e28a58f14240ffbcae47aa44413e0a : SUCCESS in 56m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d573f56adc15429caf1650fafd3d0d15 : SUCCESS in 7m 29s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e1e92128806a4bdeab0c06da56184f29 : SUCCESS in 15m 40s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/94d84788bd4f4a56ab79136074b9624f : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8d744e204fbd4a5a99d4b549ca7fa1f6 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3cbb494cdb304c16a58fff4048e03ff8 : SUCCESS in 6m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5fedd021f20a43b0b9d5abba77150639 : SUCCESS in 6m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/04c6ea5cb4f542328ec32187c0ed6326 : SUCCESS in 48m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/34b605f6074448c8b296c81f56844d55 : FAILURE in 5m 19s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b9707bfa0b6947c28a0b577a1910f773 : SUCCESS in 1h 18m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/645b1eae1bdd411d8a37d053074145a2 : FAILURE in 1h 15m 29s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/de4deba212dd4364ab6486605e2f39c2 : SUCCESS in 1h 44m 45s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8dbb366008c049f4ab62609fdf313f98 : SUCCESS in 1h 08m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/15b26654ddd14e18bebd9bc0e57e5555 : SUCCESS in 1h 37m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2a5cd6e7746a42cf86d70a27b2b6cdea : SUCCESS in 34m 38s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1db8082e2c0249cda2a0ae91715a51cd : SUCCESS in 1h 41m 12s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/397d138f3b8245bdbb3a8c60b9c78eef : SUCCESS in 1h 42m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/218b58c49eda430cb8cdaa449aeef2a3 : SUCCESS in 50m 43s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d78a0fe63c1645e39e96253076076ae2 : SUCCESS in 1h 20m 28s","accounts_in_message":[],"_revision_number":9},{"id":"605c0dee5afd135e1a04ff156c55b12ca3e30edd","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-23 18:00:29.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-05-869305-9/ns5_nfs : FAILURE in 2h 10s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-05-869305-9/ns5_iscsi : FAILURE in 2h 13s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"dc7d8eb20f640d2e6efc0bac8a4be5dc756b834b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-01-23 18:06:51.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_01_2023_17_59_15\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":9},{"id":"d2bc12d5a1188396f3d0916aa222c288691989ca","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-23 18:13:38.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/05/869305/9/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/8d7e829/ : SUCCESS in 1h 51m 58s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/05/869305/9/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/1937cbd/ : FAILURE in 2h 04m 48s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/05/869305/9/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/3105ea5/ : SUCCESS in 1h 29m 09s","accounts_in_message":[],"_revision_number":9},{"id":"410b015a0763afb8497ad8b92656fb7f27a70ff2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-23 19:04:26.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/869305/9/upstream-check/cinder-tempest-plugin-ontap-fcp/397f5f9/ : SUCCESS in 1h 05m 42s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/869305/9/upstream-check/cinder-tempest-plugin-ontap-iscsi/9a22050/ : SUCCESS in 41m 57s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/869305/9/upstream-check/cinder-tempest-plugin-ontap-nfs/8ae9865/ : SUCCESS in 43m 24s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/05/869305/9/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/69366c9/ : FAILURE in 29m 51s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"08600f25152dbd819c3b917ed393890d54c138c2","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-23 20:04:36.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/05/869305/9/ : FAILURE in 53m 49s","accounts_in_message":[],"_revision_number":9},{"id":"3576877ba55dbf8bf4c18154bbdf758af016d50d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-24 01:58:03.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/2b5fa2b6a7e049aaa9c576dbbb2729dc : POST_FAILURE in 1h 49m 25s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/61ec508ebce9455085cf073cb6ef5023 : POST_FAILURE in 1h 51m 09s","accounts_in_message":[],"_revision_number":9},{"id":"fba765d17d3cc3cb62ca767022f94b53ed58144b","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-01-24 08:16:50.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/869305-9/HPMSA-iSCSI/ : SUCCESS in 1:39:27\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/869305-9/HPMSA-FC/ : SUCCESS in 1:16:05","accounts_in_message":[],"_revision_number":9},{"id":"eec2f11ed2ed5c7a97734a661b24ba2d3b4d341a","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-01-24 10:53:44.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/869305-9/Seagate-iSCSI/ : SUCCESS in 1:57:14\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/869305-9/Seagate-FC/ : SUCCESS in 1:32:51","accounts_in_message":[],"_revision_number":9},{"id":"24c503fa411a5219dae9152da98ebd5540b5abb9","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-01-24 10:57:47.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/869305-9/PVME-iSCSI/ : SUCCESS in 1:55:30\n- PVME-FC https://os-logs.seagate.com/dell-ci/869305-9/PVME-FC/ : FAILURE in 1:35:04","accounts_in_message":[],"_revision_number":9},{"id":"f436e113d8314cc6df51f27734a4e79c9609a216","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-01-25 13:18:22.000000000","message":"Patch Set 9: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"1a1749bb2c12514a94d2742246ddec07cc8b634a","author":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"date":"2023-01-31 08:08:29.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"79e2b6c4f57b7f5c1fbe9ecee6cf30587a965045","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-02-09 19:42:13.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/869305/9/20230210041336/NEC-V-FC : SUCCESS in 27m 59s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/869305/9/20230210041336/NEC-V-ISCSI : SUCCESS in 28m 34s","accounts_in_message":[],"_revision_number":9},{"id":"8fa703e7f7585d958f9c8528d55c241c8affbf44","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-02-15 07:36:26.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/05/869305/9/20230215130140/NEC-ISCSI : SUCCESS in 3h 32m 43s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/05/869305/9/20230215130140/NEC-FC : SUCCESS in 3h 34m 43s","accounts_in_message":[],"_revision_number":9},{"id":"bcb3cbcb7fa59bb07fb95ea7e8e722b2372896af","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-02-15 08:35:41.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3eda60910e03c3a4c2a51c87e05009723a20109c","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-05-20 08:29:10.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"8c5167406f83b4365c92b47c473f63c370d0917a","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-07-22 14:13:16.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"dbe836105963f59cf4bfcc06fab4e9dc584ae266","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-07-22 14:18:32.000000000","message":"Patch Set 9:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/34/","accounts_in_message":[],"_revision_number":9},{"id":"351efebfad8df67b4379173c99b39e780a8e63bf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-07-22 14:27:03.000000000","message":"Patch Set 9:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/869305/9/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/869305/9/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"b63c90ee895ee1bc78cdcabbfcaba62fc136185c","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-07-28 11:32:41.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"36fd3109069ca23adbec2cbd1e4523d6c9011eab","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-07-28 11:33:06.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"38fc2fc1f9c8a22db3ee6aa96462f76803d64314","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-07-28 14:47:20.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b93c38892dc60634e0be8ab392ec1841a2b38aa4","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-07-28 14:47:40.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"451cc0ad911615a065cc99445bce3e9808f388af","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-07-28 14:47:49.000000000","message":"Patch Set 9:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/52/","accounts_in_message":[],"_revision_number":9},{"id":"de10680c7b40181f83e1bd1c36aa0e8bc95094da","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-07-28 14:48:18.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0b651547a499b302533632eb2d3766aa2044f835","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-07-28 14:57:31.000000000","message":"Patch Set 9:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/869305/9/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/869305/9/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":9},{"id":"ba44996178e3a4ae9613418cc6f2f2edc2a3f492","author":{"_account_id":37470,"name":"Dakshina Purohit","display_name":"Dakshina","email":"Dakshina.Purohit@ibm.com","username":"dakshina"},"date":"2025-08-18 14:57:29.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"c6345f073e15174940b184511a01b87f4025740d","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-08-18 15:28:51.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7de2740c64f846e7a4d3011c64606659365010e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"date":"2025-08-18 16:02:30.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"2ebe85aa6e64b5e432046d1aeca29c8cd60fd124","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-18 16:02:44.000000000","message":"Patch Set 10:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/200/","accounts_in_message":[],"_revision_number":10},{"id":"4a219cf911e127f3941b42a0899b9a0a04b56e9c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2025-08-18 16:16:35.000000000","message":"Patch Set 10:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/869305/10/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/869305/10/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"78f0d8fdefa9b21028f7ae9be862e87d24c71ee8","author":{"_account_id":35822,"name":"Senthil Vasudevan","display_name":"HPE AlletraMP FC CI","email":"senthil.vasudevan@hpe.com","username":"senthilv"},"date":"2025-08-18 16:28:17.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-nvmetcp-driver finger://os-nvmetcp-ci/9df4e04dad0a442e864cc5335622e736 : RETRY_LIMIT in 2s","accounts_in_message":[],"_revision_number":10},{"id":"0f094f646ddef2061b5bcaa411951c2f6fbbbd35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 19:00:50.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/296192ec18cb4e239e177ebf39d74eb1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b01bef045d646809fe29bdda8ea3f8a : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b4924e799704d37840b710b878186b5 : SUCCESS in 7m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eca9217df67c47df9854f328a81c7cc6 : SUCCESS in 7m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5ddaa1419eba40d4b8ba96deac6aee42 : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afcaac54f42544b497adcee4adb866c6 : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/70252004b3b64b4bb1e5142f3986b32b : SUCCESS in 47m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f1470d818a2547d4840ad89a367c325c : SUCCESS in 34m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aaae983f78194d4eab97af763f4d5d15 : SUCCESS in 1h 39m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0845a4802344624bcab595f6962f3f8 : SUCCESS in 34m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25d5e0eb17af4f9bbda67dd3e1787cc8 : SUCCESS in 18m 00s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/49303c3477c34bf8b597160d3fd52a18 : SUCCESS in 27m 22s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a420df5154184bd3a6fea1352309d99a : SUCCESS in 2m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a88ec317d40f4752a2e4889703dd9ddb : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dda5a336686149b28fdc8c5b2d5918f2 : SUCCESS in 5m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0443e65470104a82bde986340f5c49f4 : SUCCESS in 7m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6a2c7e236dba4cada84c00011c588ef7 : FAILURE in 11m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/08ea6ee63c4d48788365b471f18490fa : SUCCESS in 4m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9c5337688a774cb69c3661ed974dfd77 : SUCCESS in 1h 24m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f5eeb4aada204eef83ed1618771938d3 : FAILURE in 1h 46m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4e47ecfc76004052bba44898a50317d3 : SUCCESS in 1h 45m 19s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3efa6bee60204b0f819c64211ddcde6c : FAILURE in 8m 16s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/84ef82eb879b4a068df2127fa9ff67eb : SUCCESS in 20m 15s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ad3fcd07e0b64fe9acf898072a35c6d9 : SUCCESS in 1h 04m 52s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d146833f0e364a3aa4e3c9f124c3c922 : SUCCESS in 1h 30m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d81781fd14c34285b5a3ce46845a455f : SUCCESS in 24m 42s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ce0b9b22a3cf44cc9cc2cf8215aaa90f : SUCCESS in 1h 28m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7a5f42be4b954af08890eb1095384f4c : FAILURE in 12m 03s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9b480b83a61340448f51cfa210c76b6e : SUCCESS in 1h 34m 54s\n- tempest-integrated-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85cfd6e88ccb441c995b1b118a76c970 : SUCCESS in 1h 24m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc285c96801049deb2890b05e61f27fb : SUCCESS in 27m 57s","accounts_in_message":[],"_revision_number":10},{"id":"b5bde80f10d40621a9ec8fe882e76e44fe5213f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 20:39:23.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/590480fc55d245a98e51425d6563b937\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f77c5eb8414b4a9793ecf9cb054f8564 : FAILURE in 26m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c56d44dfe7b84266920cc123c3fe9a87 : SUCCESS in 27m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6af9361d38474b25a48b7c0635007f4b : SUCCESS in 18m 52s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"24bb41158044118baee8b6dec30689d56f500607","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2025-08-18 21:09:43.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/869305 : FAILURE in 1h 29m 33s","accounts_in_message":[],"_revision_number":10},{"id":"a9f5ccb40329ee80b3dd9b3607bb0af14d76b2e9","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2025-08-18 21:25:14.000000000","message":"Patch Set 10:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-05-869305-10/FC/ : UNSTABLE in 36m 45s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-05-869305-10/iSCSI/ : FAILURE in 4h 28m 28s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":10},{"id":"745fcaf795d42a018e5dfb1bca260c81e95c7172","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2025-08-18 21:28:45.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/869305 : TIMED_OUT in 3h 01m 12s","accounts_in_message":[],"_revision_number":10},{"id":"618eddfd8ac16de1299519f7eee7f06b8c11e717","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-18 21:38:05.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/359ce0826e384acfa4b507f2e97032e4 : SUCCESS in 43m 56s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/19db2f4b78e74b3c8625cd383ae9eba3 : FAILURE in 51m 10s","accounts_in_message":[],"_revision_number":10},{"id":"59de2937b202397d9be3c0a7d308dc36c69ec90d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-18 22:01:32.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1603/.","accounts_in_message":[],"_revision_number":10},{"id":"7d1f0d828b1908f9f31569f67622a311a4cfcf04","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-18 22:12:22.000000000","message":"Patch Set 10:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1468/.","accounts_in_message":[],"_revision_number":10},{"id":"4d63cc84cbf215435afabf7a03a0156e62962c03","author":{"_account_id":36348,"name":"Rahul Pawar","display_name":"Rahul Pawar","email":"rahul.p@ibm.com","username":"rahulp"},"date":"2025-08-28 12:11:04.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"57a1a9cae6565a74c645992bc5528901f05a14b7","author":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"},"date":"2025-08-29 04:35:52.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"989c0f72d28ed4875cc261926b672d7a5396fff4","author":{"_account_id":37923,"name":"anthony gamboa","display_name":"Anthony Gamboa","email":"anthony.gamboa@hitachivantara.com","username":"anthonygamboa","status":"Hitachi Vantara Software Developer"},"date":"2025-10-29 22:51:54.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1583fee76567275bced9087deb2917740db96df6","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2025-11-07 15:54:34.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"628b529b212b1eae494fc8046a48b23fb3ca929a","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-12-19 10:14:03.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e120d25b534cd7bf1c654135b6aef036c53137f2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-07-13 14:05:01.000000000","message":"Patch Set 10:\n\nBuild Started http://9.71.62.18:8080/job/CI_Run/934/","accounts_in_message":[],"_revision_number":10},{"id":"9b748e3acc212f5aea1f4600b71010ded5d05eef","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":38609,"name":"IBM Storage CI","display_name":"IBM Storage CI","email":"IBMStorageCI@wwpdl.vnet.ibm.com","username":"ibmstorageci1"},"date":"2026-07-13 14:25:18.000000000","message":"Patch Set 10:\n\nBuild Successful \n\nIBM-Storage-iSCSI-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/869305/10/iscsi.log and IBM-Storage-FC-Log: https://github.com/vp0410/IBM-Cinder-CI/tree/main/869305/10/fc.log : SUCCESS (skipped)","accounts_in_message":[],"_revision_number":10},{"id":"b9d7c18f85c43a18bcaeab281ee0afa5ad4ffe11","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-07-14 08:17:24.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"796a488014b65a0eac0eebd5a6b9ed21477a3e57","author":{"_account_id":38081,"name":"Anthony Galica","display_name":"agalica","email":"anthony.galica@hitachivantara.com","username":"agalica","status":"Hitachi Vantara"},"date":"2026-07-15 00:19:29.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"d92dd49e2a0bd7338df013443dff2287241d46c9","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-17 14:37:52.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"a10497ece92272f063a431ba75be6bb3d5a66272","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-07-17 14:38:14.000000000","message":"Patch Set 10: Code-Review-1","accounts_in_message":[],"_revision_number":10},{"id":"588f56f83cda8a120ad6fcaf3ea0091cb31b5f40","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2026-07-17 14:56:38.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0ef644543cfec446068394145630d9805a9b7a76","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2026-08-05 14:00:41.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"39da6b8e0f06dac47cc914f606af53f885ddae7a","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-08-07 14:57:40.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"8eb0b803dfc11d737c257b9b261e931cdb8a7e14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 14:58:23.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"80c38379f856c373a86fd0d6b9953d13465356c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 16:49:21.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6a1ba62f8954f948d3dbdde6daf52f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff64f2258a0448b580c007e6c1b6c8f6 : SUCCESS in 8m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a45091952e74007afb058f7f8bd2771 : SUCCESS in 13m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2fd3f8261d3f4820874a9df554fdde84 : SUCCESS in 10m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f46dc33125487cb35b5269b37d222c : SUCCESS in 18m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/0202e9d52b96455abc6ef76b34dbf4f6 : SUCCESS in 1h 02m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1cb71d5491444b57a4c92a3ce87262ef : SUCCESS in 1h 00m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a1d3c9df280a4785aa1c10ee8bcb232f : SUCCESS in 1h 32m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c768a68ed539496eb97afa4c9cb49855 : SUCCESS in 34m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f05abd1aa56f4a69b9980711934feed9 : SUCCESS in 8m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/874f166e912941e880ba9e27335c7bad : SUCCESS in 1h 12m 15s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/104b0dea21c648cfac6529a654e4d052 : SUCCESS in 1h 45m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5dd62ef115d4a2f9f7a1e84916f7012 : SUCCESS in 33m 58s","accounts_in_message":[],"_revision_number":10},{"id":"36fc888927febcaa095c6318d62d4947218b0601","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 16:49:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"1dec07d718ad165641badd46a0bd1182b29edaf5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 16:50:27.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37323125694b419886a7163c3575a483\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e64d5a1c5c84656b0aa31f6e1be956d : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d1bb75f1cdf4e1f8be1504fb3e3efc5 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"062c8c7e76cce02a09cca186c9bdce170f1cc6f7","revisions":{"5903205f11f9ef2bf276d91e509ab390ee556168":{"kind":"REWORK","_number":1,"created":"2023-01-10 06:39:14.000000000","uploader":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"ref":"refs/changes/05/869305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/869305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/869305/1"}}},"commit":{"parents":[{"commit":"489284400baec22fd948bd5a1a1a23ae89b08dce","subject":"Merge \"[SVf] As part of Flashcopy 2.0 adding support for volumegroup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/489284400baec22fd948bd5a1a1a23ae89b08dce"}]}],"author":{"name":"kumarkanishka","email":"Kumar.Kanishka@ibm.com","date":"2023-01-09 20:42:26.000000000","tz":0},"committer":{"name":"kumarkanishka","email":"Kumar.Kanishka@ibm.com","date":"2023-01-09 20:42:26.000000000","tz":0},"subject":"[SVf] lsportip needs to fetch IPs with \"host\" flag","message":"[SVf] lsportip needs to fetch IPs with \"host\" flag\n\n[Spectrum Virtualize family] During driver initialization, lsportip command\nstores the node IP details. Currently, lsportip is fetching all the IP\naddresses without any filter value. The requirement is to have the IPs for\nonly those nodes where the \"host\" flag is set to \"yes\" This will give return\nonly the IPs where the host attach-detach operations can be performed. Added\nthe \"host\" flag filtervalue in lsportip code.\n\nCloses-Bug: #1992160\nChange-Id: I067bcd66cedfac392b63c1540fd5022de24626ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5903205f11f9ef2bf276d91e509ab390ee556168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5903205f11f9ef2bf276d91e509ab390ee556168"}]},"branch":"refs/heads/master"},"11ad86a926a9b82acea81bbb7e5ddf66d40d46f8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-01-10 06:50:23.000000000","uploader":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"ref":"refs/changes/05/869305/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/869305/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/869305/2"}}},"commit":{"parents":[{"commit":"489284400baec22fd948bd5a1a1a23ae89b08dce","subject":"Merge \"[SVf] As part of Flashcopy 2.0 adding support for volumegroup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/489284400baec22fd948bd5a1a1a23ae89b08dce"}]}],"author":{"name":"kumarkanishka","email":"Kumar.Kanishka@ibm.com","date":"2023-01-09 20:42:26.000000000","tz":0},"committer":{"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","date":"2023-01-10 06:50:23.000000000","tz":0},"subject":"[SVf] lsportip needs to fetch IPs with \"host\" flag","message":"[SVf] lsportip needs to fetch IPs with \"host\" flag\n\n[Spectrum Virtualize family] During driver initialization, lsportip command stores the node IP details. Currently, lsportip is fetching all the IP addresses without any filter value. The requirement is to have the IPs for only those nodes where the \"host\" flag is set to \"yes\" This will give return only the IPs where the host attach-detach operations can be performed. Added the \"host\" flag filtervalue in lsportip code.\n\nCloses-Bug: #1992160\nChange-Id: I067bcd66cedfac392b63c1540fd5022de24626ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11ad86a926a9b82acea81bbb7e5ddf66d40d46f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11ad86a926a9b82acea81bbb7e5ddf66d40d46f8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"cc1817db20760bf87c79ac9560f7a84c20411c35":{"kind":"REWORK","_number":3,"created":"2023-01-10 10:23:27.000000000","uploader":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"ref":"refs/changes/05/869305/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/869305/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/869305/3"}}},"commit":{"parents":[{"commit":"489284400baec22fd948bd5a1a1a23ae89b08dce","subject":"Merge \"[SVf] As part of Flashcopy 2.0 adding support for volumegroup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/489284400baec22fd948bd5a1a1a23ae89b08dce"}]}],"author":{"name":"kumarkanishka","email":"Kumar.Kanishka@ibm.com","date":"2023-01-09 20:42:26.000000000","tz":0},"committer":{"name":"kumarkanishka","email":"Kumar.Kanishka@ibm.com","date":"2023-01-10 10:14:58.000000000","tz":0},"subject":"[SVf] lsportip needs to fetch IPs with \"host\" flag","message":"[SVf] lsportip needs to fetch IPs with \"host\" flag\n\n[Spectrum Virtualize family] During driver initialization,\nlsportip command stores the node IP details. Currently, lsportip is fetching\nall the IP addresses without any filter value. The requirement is to have\nthe IPs for only those nodes where the \"host\" flag is set to \"yes\" This\nwill give return only the IPs where the host attach-detach operations can\nbe performed. Added the \"host\" flag filtervalue in lsportip code.\n\nCloses-Bug: #1992160\nChange-Id: I067bcd66cedfac392b63c1540fd5022de24626ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cc1817db20760bf87c79ac9560f7a84c20411c35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cc1817db20760bf87c79ac9560f7a84c20411c35"}]},"branch":"refs/heads/master"},"731b5fa25b1d5e07b9d70836138c07f8715e1ba8":{"kind":"REWORK","_number":4,"created":"2023-01-10 10:50:25.000000000","uploader":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"ref":"refs/changes/05/869305/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/869305/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/869305/4"}}},"commit":{"parents":[{"commit":"489284400baec22fd948bd5a1a1a23ae89b08dce","subject":"Merge \"[SVf] As part of Flashcopy 2.0 adding support for volumegroup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/489284400baec22fd948bd5a1a1a23ae89b08dce"}]}],"author":{"name":"kumarkanishka","email":"Kumar.Kanishka@ibm.com","date":"2023-01-09 20:42:26.000000000","tz":0},"committer":{"name":"kumarkanishka","email":"Kumar.Kanishka@ibm.com","date":"2023-01-10 10:49:49.000000000","tz":0},"subject":"[SVf] lsportip needs to fetch IPs with \"host\" flag","message":"[SVf] lsportip needs to fetch IPs with \"host\" flag\n\n[Spectrum Virtualize family] During driver initialization,\nlsportip command stores the node IP details. Currently, lsportip\nis fetching all the IP addresses without any filter value.\nThe requirement is to have the IPs for only those nodes where the\n\"host\" flag is set to \"yes\". This will give return only the IPs\nwhere the host attach-detach operations can be performed.\nAdded the \"host\" flag filtervalue in lsportip code.\n\nCloses-Bug: #1992160\nChange-Id: I067bcd66cedfac392b63c1540fd5022de24626ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/731b5fa25b1d5e07b9d70836138c07f8715e1ba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/731b5fa25b1d5e07b9d70836138c07f8715e1ba8"}]},"branch":"refs/heads/master"},"29969e583e44ae4432011500d608440b927f6b1b":{"kind":"REWORK","_number":5,"created":"2023-01-12 07:18:15.000000000","uploader":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"ref":"refs/changes/05/869305/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/869305/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/869305/5"}}},"commit":{"parents":[{"commit":"489284400baec22fd948bd5a1a1a23ae89b08dce","subject":"Merge \"[SVf] As part of Flashcopy 2.0 adding support for volumegroup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/489284400baec22fd948bd5a1a1a23ae89b08dce"}]}],"author":{"name":"kumarkanishka","email":"Kumar.Kanishka@ibm.com","date":"2023-01-09 20:42:26.000000000","tz":0},"committer":{"name":"kumarkanishka","email":"Kumar.Kanishka@ibm.com","date":"2023-01-12 07:17:03.000000000","tz":0},"subject":"[SVf] lsportip needs to fetch IPs with \"host\" flag","message":"[SVf] lsportip needs to fetch IPs with \"host\" flag\n\n[Spectrum Virtualize family] During driver initialization,\nlsportip command stores the node IP details. Currently, lsportip\nis fetching all the IP addresses without any filter value.\nThe requirement is to have the IPs for only those nodes where the\nhost flag is set to yes. This will give return only the IPs\nwhere the host attach-detach operations can be performed.\nAdded the \"host\" flag filtervalue in lsportip code.\n\nCloses-Bug: #1992160\nChange-Id: I067bcd66cedfac392b63c1540fd5022de24626ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/29969e583e44ae4432011500d608440b927f6b1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/29969e583e44ae4432011500d608440b927f6b1b"}]},"branch":"refs/heads/master"},"80ff4b136a87bbb3a3730955e95f0fc113730d56":{"kind":"REWORK","_number":6,"created":"2023-01-13 05:22:37.000000000","uploader":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"ref":"refs/changes/05/869305/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/869305/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/869305/6"}}},"commit":{"parents":[{"commit":"489284400baec22fd948bd5a1a1a23ae89b08dce","subject":"Merge \"[SVf] As part of Flashcopy 2.0 adding support for volumegroup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/489284400baec22fd948bd5a1a1a23ae89b08dce"}]}],"author":{"name":"kumarkanishka","email":"Kumar.Kanishka@ibm.com","date":"2023-01-09 20:42:26.000000000","tz":0},"committer":{"name":"kumarkanishka","email":"Kumar.Kanishka@ibm.com","date":"2023-01-13 05:20:39.000000000","tz":0},"subject":"[SVf] lsportip needs to fetch IPs with \"host\" flag","message":"[SVf] lsportip needs to fetch IPs with \"host\" flag\n\n[Spectrum Virtualize family] During driver initialization,\nlsportip command stores the node IP details. Currently, lsportip\nis fetching all the IP addresses without any filter value.\nThe requirement is to have the IPs for only those nodes where the\nhost flag is set to yes. This will give return only the IPs\nwhere the host attach-detach operations can be performed.\nAdded the host flag filtervalue in lsportip code.\n\nCloses-Bug: #1992160\nChange-Id: I067bcd66cedfac392b63c1540fd5022de24626ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/80ff4b136a87bbb3a3730955e95f0fc113730d56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/80ff4b136a87bbb3a3730955e95f0fc113730d56"}]},"branch":"refs/heads/master"},"486a13a421dbe9be2442e9d97a71061056dee7c5":{"kind":"REWORK","_number":7,"created":"2023-01-16 08:06:24.000000000","uploader":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"ref":"refs/changes/05/869305/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/869305/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/869305/7"}}},"commit":{"parents":[{"commit":"489284400baec22fd948bd5a1a1a23ae89b08dce","subject":"Merge \"[SVf] As part of Flashcopy 2.0 adding support for volumegroup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/489284400baec22fd948bd5a1a1a23ae89b08dce"}]}],"author":{"name":"kumarkanishka","email":"Kumar.Kanishka@ibm.com","date":"2023-01-09 20:42:26.000000000","tz":0},"committer":{"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","date":"2023-01-16 08:06:24.000000000","tz":0},"subject":"[SVf] lsportip needs to fetch IPs with \"host\" flag","message":"[SVf] lsportip needs to fetch IPs with \"host\" flag\n\n[Spectrum Virtualize family] During driver initialization,\nlsportip command stores the node IP details. Currently, lsportip\nis fetching all the IP addresses without any filter value.\nThe requirement is to have the IPs for only those nodes where the\nhost flag is set to yes. This will give return only the IPs\nwhere the host attach-detach operations can be performed.\nAdded the host flag filtervalue in lsportip code.\n\nCloses-Bug: #1992160\nChange-Id: I067bcd66cedfac392b63c1540fd5022de24626ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/486a13a421dbe9be2442e9d97a71061056dee7c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/486a13a421dbe9be2442e9d97a71061056dee7c5"}]},"branch":"refs/heads/master"},"58032177256123f9102579150b1f1d86457d0c93":{"kind":"REWORK","_number":8,"created":"2023-01-20 05:35:40.000000000","uploader":{"_account_id":35193,"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","username":"kumarkanishka"},"ref":"refs/changes/05/869305/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/869305/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/869305/8"}}},"commit":{"parents":[{"commit":"489284400baec22fd948bd5a1a1a23ae89b08dce","subject":"Merge \"[SVf] As part of Flashcopy 2.0 adding support for volumegroup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/489284400baec22fd948bd5a1a1a23ae89b08dce"}]}],"author":{"name":"kumarkanishka","email":"Kumar.Kanishka@ibm.com","date":"2023-01-09 20:42:26.000000000","tz":0},"committer":{"name":"Kumar Kanishka","email":"Kumar.Kanishka@ibm.com","date":"2023-01-20 05:35:40.000000000","tz":0},"subject":"[SVf] lsportip needs to fetch IPs with \"host\" flag","message":"[SVf] lsportip needs to fetch IPs with \"host\" flag\n\n[Spectrum Virtualize family] During driver initialization,\nlsportip command stores the node IP details. Currently, lsportip\nis fetching all the IP addresses without any filter value.\nThe requirement is to have the IPs for only those nodes where the\nhost flag is set to yes. This will give return only the IPs\nwhere the host attach-detach operations can be performed.\nAdded the host flag filtervalue in lsportip code.\n\nCloses-Bug: #1992160\nChange-Id: I067bcd66cedfac392b63c1540fd5022de24626ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/58032177256123f9102579150b1f1d86457d0c93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/58032177256123f9102579150b1f1d86457d0c93"}]},"branch":"refs/heads/master"},"b9a05ea08a5b1ed93973580264eb02cd3008d9ec":{"kind":"REWORK","_number":9,"created":"2023-01-23 15:59:46.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/05/869305/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/869305/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/869305/9"}}},"commit":{"parents":[{"commit":"489284400baec22fd948bd5a1a1a23ae89b08dce","subject":"Merge \"[SVf] As part of Flashcopy 2.0 adding support for volumegroup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/489284400baec22fd948bd5a1a1a23ae89b08dce"}]}],"author":{"name":"kumarkanishka","email":"Kumar.Kanishka@ibm.com","date":"2023-01-09 20:42:26.000000000","tz":0},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2023-01-23 15:59:29.000000000","tz":-300},"subject":"[SVf] lsportip needs to fetch IPs with \"host\" flag","message":"[SVf] lsportip needs to fetch IPs with \"host\" flag\n\n[Spectrum Virtualize family] During driver initialization,\nlsportip command stores the node IP details. Currently, lsportip\nis fetching all the IP addresses without any filter value.\nThe requirement is to have the IPs for only those nodes where the\nhost flag is set to yes. This will give return only the IPs\nwhere the host attach-detach operations can be performed.\nAdded the host flag filtervalue in lsportip code.\n\nCloses-Bug: #1992160\nChange-Id: I067bcd66cedfac392b63c1540fd5022de24626ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b9a05ea08a5b1ed93973580264eb02cd3008d9ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b9a05ea08a5b1ed93973580264eb02cd3008d9ec"}]},"branch":"refs/heads/master"},"062c8c7e76cce02a09cca186c9bdce170f1cc6f7":{"kind":"REWORK","_number":10,"created":"2025-08-18 16:02:30.000000000","uploader":{"_account_id":37328,"name":"Nimesh Desai","display_name":"Nimesh Desai","email":"nimesh.desai@ibm.com","username":"nimeshdesai"},"ref":"refs/changes/05/869305/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/05/869305/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/05/869305/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/05/869305/10"}}},"commit":{"parents":[{"commit":"2a6c3565438554fd857ddbb57bb65ca05a17d91b","subject":"Merge \"NetApp - Fixed detach issue for multi-attached volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2a6c3565438554fd857ddbb57bb65ca05a17d91b"}]}],"author":{"name":"kumarkanishka","email":"Kumar.Kanishka@ibm.com","date":"2023-01-09 20:42:26.000000000","tz":0},"committer":{"name":"nimeshdesai","email":"Nimesh.Desai@ibm.com","date":"2025-08-18 16:02:15.000000000","tz":330},"subject":"[SVf] lsportip needs to fetch IPs with \"host\" flag","message":"[SVf] lsportip needs to fetch IPs with \"host\" flag\n\n[Spectrum Virtualize family] During driver initialization,\nlsportip command stores the node IP details. Currently, lsportip\nis fetching all the IP addresses without any filter value.\nThe requirement is to have the IPs for only those nodes where the\nhost flag is set to yes. This will give return only the IPs\nwhere the host attach-detach operations can be performed.\nAdded the host flag filtervalue in lsportip code.\n\nCloses-Bug: #1992160\nChange-Id: I067bcd66cedfac392b63c1540fd5022de24626ac\nSigned-off-by: nimeshdesai \u003cNimesh.Desai@ibm.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/062c8c7e76cce02a09cca186c9bdce170f1cc6f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/062c8c7e76cce02a09cca186c9bdce170f1cc6f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38067,"name":"Sandip Rajbanshi","display_name":"Sandip27","email":"sandip.rajbanshi@ibm.com","username":"sandip-rajbanshi-27"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
