)]}'
{"id":"openstack%2Fcinder~869981","triplet_id":"openstack%2Fcinder~stable%2Fzed~Ia0353204abf849467106ee08982d1271de23101a","project":"openstack/cinder","branch":"stable/zed","topic":"qemu-img-info-json","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-01-13 15:55:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia0353204abf849467106ee08982d1271de23101a","subject":"Use the json format output of qemu-img info","status":"MERGED","created":"2023-01-12 13:47:41.000000000","updated":"2023-01-17 02:04:09.000000000","submitted":"2023-01-13 15:55:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869981-qemu-img-info-json","meta_rev_id":"d28e3fa89e259ea29923c75ad99c4ebdb804f9f9","_number":869981,"virtual_id_number":869981,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-13 15:55:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-13 13:28:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":2,"date":"2023-01-13 14:33:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":1,"date":"2023-01-13 14:33:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}],"REVIEWER":[{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-12 13:47:41.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-01-12 14:07:07.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-01-12 14:47:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-12 15:42:29.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-12 16:03:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-12 18:43:50.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-01-12 23:18:12.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-01-13 00:20:31.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-13 13:28:16.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-01-13 14:33:21.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2023-01-17 02:04:09.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"fb64a07a61dbdf00a046f1e30f8a33c1798bfd19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-01-12 13:47:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"811b5a061c85a78217c2836814135a469b7907cc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-12 14:07:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"58870b5ca3a63ee71ba626491d39cc1917059ffd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:47:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b376b52f6b047dcb005af5d80f9f07f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6d4f0355e6b42c880a52a201ef362dc : SUCCESS in 27m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d3dcaa37a0954b19a9735604f9726169 : SUCCESS in 25m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"17b83c07fea08461dd46017bc459f9ec3b9cfa58","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-12 15:42:29.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-81-869981-1/ns5_nfs : FAILURE in 1m 34s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-81-869981-1/ns5_iscsi : FAILURE in 1m 28s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"226bd25547060d5057a39b23c5aa9ca95fca4d37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 16:03:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40015cbc7d304114b684acc4368ea4be\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a29ddf631d14b03a633a88824fadf24 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9802011534748688e37fe63f2b0fd4b : SUCCESS in 10m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f5e64a3126e4b77a04e4ec76c6823c1 : SUCCESS in 10m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91ca62581edc41569f40e1afa82b9a02 : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d07e92563d6b4ed8b08b780e22beaa44 : SUCCESS in 12m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/0fa8fd6542124eb8ba84d553d82d049d : SUCCESS in 56m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4731726532fe4710a629f9968a74363d : SUCCESS in 1h 32m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d276cffca03d4f0591f744d95fc6ac1e : SUCCESS in 1h 05m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/31ad603018b74a0992244841e0b78b18 : SUCCESS in 14m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e999cc65747d43c3acd1f99c4189fa1c : SUCCESS in 5m 19s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/076ebbf2ffae49cdb52d35bfedf70585 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/76301f659f634f918d7f87fb6ae57fbd : SUCCESS in 5m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/448808d6b86a46d0afd5ee5c8f57775b : SUCCESS in 6m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d2b3034f91a24ac0849ac7f3712c4b8f : SUCCESS in 46m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/21e8ab5fa5ef4cb89b94fc818d0fcf24 : FAILURE in 6m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c13ec13fbaf444e7a68ec8c276c0adab : SUCCESS in 1h 27m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/21b941eb9cae44b8bab12d00ad979795 : FAILURE in 1h 40m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d9a951d6fa7746cb8360b5f8f2af41ee : SUCCESS in 1h 36m 40s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/559f3f0399f44b0d85f5eda8aaa2f827 : SUCCESS in 1h 19m 56s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d3c4d235a2c24e2a8aabae6a0f41684e : SUCCESS in 1h 33m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/86cc3dae38d640cfb6f2a9371499a926 : SUCCESS in 46m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/bf7b69757e3441b588852a70490c7e5e : SUCCESS in 1h 19m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fd5ca89112b344309a9f3b59af92b59d : SUCCESS in 2h 13m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5106ed099bff4e30ba74e1de239ae72b : SUCCESS in 1h 06m 19s","accounts_in_message":[],"_revision_number":1},{"id":"3dda4551b1592441993eb822ad295b1f4367bafb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-12 18:43:50.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"197b7d3769a3d58879d816ac6e0c206a94ef46ba","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-12 23:18:12.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/81/869981/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/bc8602d/ : SUCCESS in 1h 40m 26s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/81/869981/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/cba1c7f/ : FAILURE in 2h 39m 29s\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/81/869981/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/73508fe/ : FAILURE in 1h 46m 15s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy pure-devstack-cinder-tempest-fc-aio-jammy : NODE_FAILURE in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/81/869981/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/eebb8bd/ : FAILURE in 2h 06m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d1a6484805e883704def0322a8aa799f90fe4497","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-13 00:20:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/869981 : FAILURE in 1h 21m 49s","accounts_in_message":[],"_revision_number":1},{"id":"251ded707b586559f8bef0adfe6a4d26fe6fe983","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-13 02:57:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/869981 : FAILURE in 1h 07m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e0acd2db404d2c7ab6e528dfbcdaf8648a2b5fef","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-13 06:23:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/81/869981/1/check/iscsi-driver-nimble/a973a01/a973a01.html : FAILURE in 1h 25m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b02594b8c60dd9e9e1398e7081fbbe5feb9f8cc1","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-01-13 13:28:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e3625eb650fa8604a8f68b597f1bff4528443e81","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2023-01-13 14:33:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2d56b761d1abccc909869134e188ccb86655dfa1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 14:34:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42f63cbee13e5bad0be6b5cbda23434f230a022d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 15:55:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a166f02d2fc49b6bdc48777a7ae261f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce37af42ee0749bc889df873975f7455 : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acdb1d9b4cc444cb94bb871ad48f7b50 : SUCCESS in 9m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ce62d9c787b46ceb6fb810dca049add : SUCCESS in 12m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab8b5975fe4e4d06926c8e1bce75f268 : SUCCESS in 11m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/0620e2832de4401f849b7218cd109d75 : SUCCESS in 1h 03m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1908d2401c3b4732a9977a1ddb06dbc8 : SUCCESS in 1h 17m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b6ddec7e02b4d4889801c1cfe38fc88 : SUCCESS in 54m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/75bbe6bf59024d47a4881803b68b529b : SUCCESS in 1h 13m 09s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a386faf75fe74c4b951b3f26db927870 : SUCCESS in 1h 15m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7f05ba3cf6b4e47b50b165e783e7fd0 : SUCCESS in 51m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d17944a118eb8d6979d41ff484855d9e5ec8ce6a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 15:55:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0fe9e074331282fda663cfa8fa4aace30efe32e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 15:56:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e01b9eeeb384364b50e532a94499883\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bc35d4f22c542699156c8baff20ac70 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1},{"id":"d28e3fa89e259ea29923c75ad99c4ebdb804f9f9","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-17 02:04:09.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/81/869981/1/ : FAILURE in 48m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11b0f97a013189fe985deec4f776a7719a6ad089","revisions":{"11b0f97a013189fe985deec4f776a7719a6ad089":{"kind":"REWORK","_number":1,"created":"2023-01-12 13:47:41.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/81/869981/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/81/869981/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/869981/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/869981/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/81/869981/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/81/869981/1"}}},"commit":{"parents":[{"commit":"f1ebe259df3e81a241a708ecd50b6e550ac108ea","subject":"Revert \"PowerFlex driver - fix the display of the incorrect volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f1ebe259df3e81a241a708ecd50b6e550ac108ea"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-06 12:48:44.000000000","tz":540},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-01-12 13:41:43.000000000","tz":-300},"subject":"Use the json format output of qemu-img info","message":"Use the json format output of qemu-img info\n\nSupport for the human format by oslo_utils.imageutils.QemuImgInfo was\ndeprecated since oslo.utils 4.9.1 [1]. This change replaces the human\nformat with the json format which will be used by default.\n\n[1] 73eb0673f627aad382e08a816191b637af436465\n\nBackport note: The json format is preferable because it allows access\nto format specific details (since oslo.utils 4.1.0).  These details\nare not present when the default \u0027human\u0027 format is used.  See change\nI133da07a5a9628b8a9 for details.\n\nCloses-Bug: #1940540\nChange-Id: Ia0353204abf849467106ee08982d1271de23101a\n(cherry picked from commit c0d2e7ebd8b992d61a5f486baad3016d79e6a5d9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11b0f97a013189fe985deec4f776a7719a6ad089"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/11b0f97a013189fe985deec4f776a7719a6ad089"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
