)]}'
{"id":"openstack%2Fcinder~870550","triplet_id":"openstack%2Fcinder~master~I6e1a9026e677aee6c7ccad908fe6f92dc253762a","project":"openstack/cinder","branch":"master","topic":"bug/2002996","attention_set":{},"removed_from_attention_set":{"33473":{"account":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"last_update":"2024-05-30 11:42:38.000000000","reason":"Change was submitted"},"35075":{"account":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"last_update":"2024-01-19 14:12:02.000000000","reason":"\u003cGERRIT_ACCOUNT_35075\u003e replied on the change","reason_account":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"}},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2024-05-30 11:42:38.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2024-05-30 11:42:38.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-05-29 17:25:14.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},"12988":{"account":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"last_update":"2024-05-30 11:42:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6e1a9026e677aee6c7ccad908fe6f92dc253762a","subject":"StorPool: drop copy_image_to_volume() and copy_volume_to_image()","status":"MERGED","created":"2023-01-16 14:56:42.000000000","updated":"2024-06-02 19:35:58.000000000","submitted":"2024-05-30 11:42:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870550-bug/2002996","meta_rev_id":"d2ca155a0beb64b839cb179910fc3eadc1bb5ad5","_number":870550,"virtual_id_number":870550,"owner":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-30 11:42:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":2,"date":"2024-05-29 14:21:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-05-29 14:21:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":2,"date":"2024-05-29 17:25:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2024-05-29 17:25:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-16 15:02:59.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-16 16:12:12.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-01-16 17:02:05.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-01-16 17:27:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-16 17:52:13.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-16 18:18:33.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-16 21:08:12.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-17 00:07:25.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-01-17 00:12:13.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-17 04:46:31.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-01-17 06:00:39.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-17 08:29:11.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-01-18 12:53:31.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-01-19 09:03:36.000000000","updated_by":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"reviewer":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"state":"REVIEWER"},{"updated":"2023-01-19 11:26:37.000000000","updated_by":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"reviewer":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"state":"REVIEWER"},{"updated":"2023-01-21 12:56:46.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-26 12:35:23.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-01-26 13:16:31.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-01-26 15:10:57.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-01-26 17:05:05.000000000","updated_by":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"reviewer":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"state":"CC"},{"updated":"2023-01-27 15:42:27.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-01-29 12:27:50.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2023-01-29 12:31:44.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-03-02 12:36:51.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-03-02 16:49:17.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-03-02 18:07:09.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2023-03-08 22:09:27.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2023-05-09 01:12:29.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-05-09 18:54:54.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-18 12:57:15.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-19 11:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-02-02 02:46:37.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-04-22 14:41:56.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-05-29 10:45:50.000000000","updated_by":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"reviewer":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"state":"CC"},{"updated":"2024-05-30 03:39:54.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"CC"},{"updated":"2024-06-01 20:50:56.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"}],"messages":[{"id":"3ba2ff4e117f333f1a7233fefcf496842887f6c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-01-16 14:56:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"142cd608717cf5cdf6da70ec3436f092ca490117","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-16 15:02:59.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7685815ca902fb16e9bf7ecde8e36b898facfe01","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-16 16:12:12.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"198a43fc527e9085a0e1c96e6eb34d1830e0e661","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-16 17:02:05.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/50/870550/1/check/iscsi-driver-nimble/38c46c9/38c46c9.html : FAILURE in 14m 37s","accounts_in_message":[],"_revision_number":1},{"id":"cd30de8cef5bc9475b36d8048d2937c7c4e98800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 17:27:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c89893efd9694f199e74136a260c2fb7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d75c256c50743a3ac2b565ebe93c606 : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa187f12d4594182b2b6c04b50107713 : SUCCESS in 10m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36c34dc0a1344212bdb1037bf19bbc24 : SUCCESS in 11m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e1b597bf13c4216933f20dc4054d640 : SUCCESS in 11m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/b8b21ae9655c44f9bf62ddc117faa3f4 : SUCCESS in 44m 29s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fa8c053529664a25aedf309035c33c5b : SUCCESS in 49m 16s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/08ffc3bc972d421a8d2666fd61fc886c : TIMED_OUT in 2h 06m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce5eedf800844ecbb8fd3466625fdb3f : SUCCESS in 44m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fd065727edb8411086a48dbeccc975e1 : SUCCESS in 14m 48s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/3ab9c12008a74bdf94f1f87c75908465 : SUCCESS in 5m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2149d81cd8fb4b258352650bd75b1ebb : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d306c2b9356646279931667577ca46db : SUCCESS in 5m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f885aa75a8414dbaa4c657aa85b21ade : SUCCESS in 5m 26s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/519bbad657d44b84b0f180dc620bc876 : SUCCESS in 38m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/662da6dce3a849ea8c9e7a2e1dd73e0e : FAILURE in 4m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2db6fdc7f0724e45b76d4af700133986 : SUCCESS in 1h 13m 25s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bb405cf2c7d2484ea28ab81bf3b313b2 : FAILURE in 1h 16m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d5b27f1b306d420296e69cfc01e4316e : SUCCESS in 1h 32m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f52edff20e024a0a8d90ccc335b69b7b : SUCCESS in 1h 05m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6873a2469e17490dab914b46dd7e9ab2 : SUCCESS in 1h 39m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c9141050c6b645e2a11e2a2a6d985b43 : SUCCESS in 30m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/292688dce2b34b65ae0d68171c19280d : SUCCESS in 1h 28m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bd28a84a3bac41a092b9a14019a48f93 : SUCCESS in 2h 11m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69ebec28ee8f47e883604039db037b89 : SUCCESS in 42m 53s","accounts_in_message":[],"_revision_number":1},{"id":"fa937a757196fd63fdfedef6ee2a978f635ab31b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-16 17:52:13.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-50-870550-1/ns5_nfs : FAILURE in 39m 40s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-50-870550-1/ns5_iscsi : FAILURE in 39m 11s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"04d067e2eff58d6bc2849ccdcb19fc21bca67fa8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-16 17:55:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/50/870550/1/check/iscsi-driver-alletra-6k/9488fa1/9488fa1.html : FAILURE in 26m 37s","accounts_in_message":[],"_revision_number":1},{"id":"0f69f38079743826223e57a6f32f275270a3f934","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-16 18:18:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/870550 : FAILURE in 23m 24s","accounts_in_message":[],"_revision_number":1},{"id":"2f7d25fa3ca0a3a7e9abff0c806250fc7eec29aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 18:19:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd0a801f9930488a865a5675b198841b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d9697ce0ca24aba83104d18a9d96923 : SUCCESS in 23m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/77bea31f48654279ac1c1f2696bc282f : FAILURE in 28m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9bc45002243087f060e0e4d8ffa7bb7fbc57e8f","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-16 18:25:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/870550 : FAILURE in 26m 06s","accounts_in_message":[],"_revision_number":1},{"id":"030896a01b7f8b73a8145835cd500c78232fb3da","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-16 21:08:12.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/1e3a9b59256641bea62dee0262592362 : POST_FAILURE in 2h 29m 00s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/410f90c381f347bf969de5e182efe718 : POST_FAILURE in 2h 07m 25s","accounts_in_message":[],"_revision_number":1},{"id":"589c2ca6fcaa4da86817dc2461054efd01bf58d1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-17 00:07:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/50/870550/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/91f0a9c/ : SUCCESS in 1h 49m 19s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/50/870550/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6906b16/ : FAILURE in 2h 24m 59s\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/50/870550/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/743eb75/ : SUCCESS in 2h 00m 07s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy pure-devstack-cinder-tempest-fc-aio-jammy : NODE_FAILURE in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/50/870550/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/ee8be89/ : SUCCESS in 1h 37m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e4311550a07bf88105eb771bef35638501960c43","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-17 00:12:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/50/870550/1/ : FAILURE in 44m 54s","accounts_in_message":[],"_revision_number":1},{"id":"aa41276d37a14d902cb11e57dbdb3cff90b4190b","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-01-17 01:41:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ada1aa3f0fd0bbbcb37538ac3c06f916e85e539","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-17 01:54:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"20e50bbe1cca4d126d7f993198ac06e072a28687","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 02:14:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cce5e96e67db40db8e896fecb6001c53\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a35877fdbc54492b9d8ce6fe8cb5243 : FAILURE in 28m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e637751f14db4009b1d60094af01398d : FAILURE in 30m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a939ba5fb7aa239485f0b6c59b1da29a3deecedb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-17 02:49:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d385c6b371c84948b87b0403160e3f7a : RETRY_LIMIT in 8m 39s","accounts_in_message":[],"_revision_number":1},{"id":"282e8faa475d3c752222e83ecd7b7c414b0a716b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 03:48:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fcf6bcab87943fe93293cd01abe5f2e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38d4d8be26a74612815096b8e4b6933f : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55ac1135614f402895e77f2267de3801 : SUCCESS in 10m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2df86ec72484e2091090175576f3267 : SUCCESS in 11m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a2f3335f6224769b0c19d69e24f5163 : SUCCESS in 12m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/f21e705423b84d24bdeb3d1011052691 : SUCCESS in 44m 45s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5c899bd9a1ff43cf8983dcb7921c03b8 : SUCCESS in 58m 10s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3cbd84d6cd6e4d2985e652f4f0c63e9f : SUCCESS in 1h 20m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/408d11b7f3a1404492d17506cf32f554 : SUCCESS in 1h 12m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/50ce8543544d448fbc10f37c5b573622 : SUCCESS in 14m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c7905b9ed685468393809cb8bb91a5c4 : SUCCESS in 6m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/09ebf07f7cb145848ad5f2eb6abf8c78 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2449f97e82ac408f8e619b2d65f3f454 : SUCCESS in 5m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ec6d6342e41a44aa8a0b85bbbdb1858b : SUCCESS in 6m 34s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/69f733a21ae845eb9c6a44464e6cd341 : SUCCESS in 38m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c82b904ddd1c46b79fcd191b7d73c522 : FAILURE in 5m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b06c122b9561403485e44d01ab3e4a33 : SUCCESS in 1h 08m 12s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6ec92e087c5b490a8f863e05a61efc74 : FAILURE in 1h 18m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6e1fa8b1dae0477c8a5d20e01c53cd3b : SUCCESS in 1h 30m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e50c9703111c4ea39202b06665cbe4e9 : SUCCESS in 1h 23m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0594212262ee404cb45bf38540876a7a : SUCCESS in 1h 27m 25s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fea51620d203441aaee54fa1b1dafc13 : SUCCESS in 33m 05s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/df195b39996c4fd2b24afafef513ef40 : SUCCESS in 1h 22m 34s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2f5563930b234c5abb8ed3809009f3af : SUCCESS in 2h 03m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa383d95647f483fb25cd945a341546e : SUCCESS in 47m 24s","accounts_in_message":[],"_revision_number":1},{"id":"99e8a88e94a7d93ff8e292d653c5cfa43c85518d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-17 04:46:31.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/1/upstream-check/cinder-tempest-plugin-ontap-fcp/cd67f1f/ : SUCCESS in 56m 12s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/19220b6/ : SUCCESS in 41m 52s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/1/upstream-check/cinder-tempest-plugin-ontap-nfs/5d1e2e5/ : SUCCESS in 43m 15s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"d1964883bc34bf24ad031defa47e853d1cff9b38","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-17 05:30:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/50/870550/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/57ebb9a/ : SUCCESS in 1h 51m 37s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/50/870550/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/26f1a08/ : FAILURE in 2h 36m 47s\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/50/870550/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/2dc4f2c/ : SUCCESS in 2h 04m 08s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/50/870550/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/fe6a779/ : FAILURE in 2h 10m 16s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/50/870550/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/0fcda99/ : SUCCESS in 1h 55m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2322ceaa8e0eecbd5d3bec844858db13bab885a7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-01-17 06:00:39.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/50/870550/1/check/DellEMC_PowerFlex/e59e20c/DellEMC_PowerFlex/1530 : SUCCESS in 1h 30m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"bbec5f5200d0b235a0cb5fba1be21f279da74db8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-17 07:21:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/92d7e446627d4c61be0dbf03900fcb3d : RETRY_LIMIT in 8m 55s","accounts_in_message":[],"_revision_number":1},{"id":"2ad6c4212faf0ffeb5ed1d9707aa83994360c293","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-17 08:29:11.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/50/870550/1/check/DellEMC_Unity_FC/a45374a/DellEMC_Unity_FC/2160 : SUCCESS in 1h 36m 21s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/50/870550/1/check/DellEMC_Unity_iSCSI/863c1ee/DellEMC_Unity_iSCSI/2029 : FAILURE in 2h 46m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"01a7a5d8355e5a36f9a5c368f2b7ead0713da807","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-01-18 12:53:31.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/1/20230118150243/NEC-V-FC : SUCCESS in 6h 00m 45s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/1/20230118150243/NEC-V-ISCSI : SUCCESS in 6h 50m 43s","accounts_in_message":[],"_revision_number":1},{"id":"3345fde38362380624611642e5c7b240c77d0467","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-01-19 09:03:36.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"feac0772550a7084955817abef116e0611aaf784","author":{"_account_id":33473,"name":"Atsushi Kawai","display_name":"akawai","email":"atsushi.kawai.bu@hitachi.com","username":"akawai"},"date":"2023-01-19 11:26:37.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e37e5e04e9296453f10e249c2f1d71c9a60d982b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-01-21 12:56:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/1/20230121151644/NEC-ISCSI : SUCCESS in 3h 36m 37s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/1/20230121151644/NEC-FC : SUCCESS in 6h 39m 59s","accounts_in_message":[],"_revision_number":1},{"id":"3268ff389e5a87f749fda9e99a4b3d266cfcd82b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-01-26 12:35:12.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"96c643bc5258296db24c90bb665161c68f735824","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-01-26 12:35:23.000000000","message":"Patch Set 2:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2},{"id":"3d60108362b50c6a6fbb02d1216e8ed8f20c9770","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-26 12:41:15.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"07dbac2cc45f839e94332ec83f514930984286a2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-26 12:47:57.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"643dadc0086694989cb369f1045586d51778082b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 12:52:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad262a7677aa43d2a5b8b4d04fb7b48b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/62987875cee941af9cd00621a5671d53 : SUCCESS in 14m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/68a28c4274dc43bca8c5fda13b31c7cb : SUCCESS in 12m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"005fba69a317953580625f12961dab9cdd226364","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-01-26 13:16:31.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/44c7e77794b24660b2653ef1ee8b5780 : POST_FAILURE in 8m 11s","accounts_in_message":[],"_revision_number":2},{"id":"532537fa317d8272d95f63b2c24e1f926f08a0de","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-26 14:46:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/870550 : FAILURE in 23m 28s","accounts_in_message":[],"_revision_number":2},{"id":"dbcb5431b7db6f1e8d9780a9933a28973120113b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-26 14:49:22.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/50/870550/2/check/iscsi-driver-alletra-6k/6cf37eb/6cf37eb.html : FAILURE in 26m 01s","accounts_in_message":[],"_revision_number":2},{"id":"d7755433d77c0c298a7a1f8e572dcc550c0950c9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-26 14:49:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/870550 : FAILURE in 26m 07s","accounts_in_message":[],"_revision_number":2},{"id":"7627a21a74d86bcc77432491923f43b7d29fc98a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-26 14:57:11.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6140cb5855114d0296fc6bfdd824739d : SUCCESS in 1h 08m 13s","accounts_in_message":[],"_revision_number":2},{"id":"0c3abcd33046a7a180935f71618e383207ffd33c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 15:03:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9314ed10dad14ffea188b329129c2b35\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/648064169a08437c866e5826325c8f52 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9e597c252d94a32b22b373fff2d9189 : SUCCESS in 9m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f2bb8199a7f4a46bc9a7759959663b6 : SUCCESS in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfea239b8afd464c8f1811cd2ca4bee6 : SUCCESS in 12m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/64e3ccfd4dcd4c309a837ad19f625b38 : SUCCESS in 59m 50s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a11828835158469398600040a52dcd67 : SUCCESS in 1h 01m 18s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/23bc71215ff340e3b0d30db644387c1d : SUCCESS in 1h 30m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3b7818a34d042ba9c23ba3cc64a922e : SUCCESS in 56m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6fbb74bc1850405cbccf450d6e21b4e6 : SUCCESS in 16m 21s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/00fc310679c348a2a08173d4b8615034 : SUCCESS in 5m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8e1b0c9f8457470099d75a01ebd8f7f5 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1e2367bed78f45d0b4c56ec6d2de092a : SUCCESS in 6m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c8cf83f2dbee4ec1a20f36189b911541 : SUCCESS in 6m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3ab81708c03e456da34bf742bc0f5e39 : SUCCESS in 36m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2e222d38278e434e84886d961a3467a8 : FAILURE in 4m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9b0cc407f75c42e2a7f9dcd085302bd9 : SUCCESS in 1h 03m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/66730233242142099426c42fd6124113 : FAILURE in 1h 27m 03s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ef6218fd72974df79ab3549e5a8abd1e : SUCCESS in 1h 49m 54s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/42cb982b062448d69d53caf2b11e0339 : SUCCESS in 1h 01m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/247d6993694544f3b5c8eff9353fa133 : SUCCESS in 1h 40m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4a2b98439d6b4380bf6eb25c64eace38 : SUCCESS in 40m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/33f361b37c3e49ab885e3caafddd6668 : SUCCESS in 1h 44m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/97e4637c411041d690858788aefac933 : SUCCESS in 2h 05m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70b45cc82d4b44ec96962b68d94f6c26 : SUCCESS in 53m 10s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46b19328c8ba4bc19eabdef577569ba5 : SUCCESS in 1h 23m 32s","accounts_in_message":[],"_revision_number":2},{"id":"a95d6acffb079592857a09f1b279ce5a2d960c89","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-01-26 15:10:57.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/26_01_2023_15_03_13\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"c105a4a2aa1566ea703930bafec49830245ea1c3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-26 16:41:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/2/upstream-check/cinder-tempest-plugin-ontap-fcp/6314894/ : SUCCESS in 1h 36m 35s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/8050ce0/ : SUCCESS in 47m 58s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/2/upstream-check/cinder-tempest-plugin-ontap-nfs/99561f2/ : SUCCESS in 46m 16s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/20e9e1b/ : FAILURE in 33m 03s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"e12ba5ef7c82f2ae3c53a943f0d5dbe8c9b6061c","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-01-26 17:05:05.000000000","message":"Patch Set 2:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/50/870550/2/fc : FAILURE in h m 0s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/50/870550/2/iscsi : FAILURE in h m 0s","accounts_in_message":[],"_revision_number":2},{"id":"0290f935afa9bd68ebf5d0f778eb192a0c85e220","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-27 02:21:26.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-50-870550-2/ns5_nfs : FAILURE in 11h 49m 45s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-50-870550-2/ns5_iscsi : FAILURE in 12h 15m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"693f45afcc78f1cb3fb2199e0a04019ccde2ae24","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-27 05:36:50.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/66231b93fe724c7cb80f57cc324f14e7 : POST_FAILURE in 2h 09m 15s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/b9bf77a043ae44738c8946d921c559ac : POST_FAILURE in 1h 32m 31s","accounts_in_message":[],"_revision_number":2},{"id":"d5e5beac754ba3da9dab72b6dd037fba8f7e4acd","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-27 05:50:37.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/50/870550/2/ : FAILURE in 1h 00m 33s","accounts_in_message":[],"_revision_number":2},{"id":"b90f5677a89b7fff7634c827f7c0dc584af121af","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-01-27 15:42:27.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/870550-2/HPMSA-iSCSI/ : FAILURE in 1:40:57\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/870550-2/HPMSA-FC/ : SUCCESS in 1:17:16","accounts_in_message":[],"_revision_number":2},{"id":"c9124397103685efcdd29f63590ad0274bd281be","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-01-29 12:27:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/870550-2/Seagate-iSCSI/ : SUCCESS in 1:57:09\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/870550-2/Seagate-FC/ : SUCCESS in 1:33:16","accounts_in_message":[],"_revision_number":2},{"id":"8e7dd7dbda77dccf18bc4e1df26c89cfc50778c7","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-01-29 12:31:44.000000000","message":"Patch Set 2:\n\nBuild failed.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/870550-2/PVME-iSCSI/ : FAILURE in 2:00:32\n- PVME-FC https://os-logs.seagate.com/dell-ci/870550-2/PVME-FC/ : FAILURE in 1:34:38","accounts_in_message":[],"_revision_number":2},{"id":"1c9d4ca32fdc7a775d3f9050790dfc038f15e550","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-02-08 13:33:35.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"926b2704e5c65973776830b86c9f6fab1c8f6756","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-08 13:49:14.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"5becca962c9360e498eb5cf4c8937f70f4b25139","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 13:49:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9882c7e99cb84fe2a5f2b15c9c7733aa\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ba87e32078e4a7096ca76e1f3228a2b : SUCCESS in 12m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9457df109b5043f18623e6817379f1bb : SUCCESS in 11m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"60ed47bb8a9892f3a87fd10a7a393d31cbd51c2c","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-08 15:05:09.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/6e367e62d0fd4922b4b98549501a7846 : POST_FAILURE in 12m 45s","accounts_in_message":[],"_revision_number":3},{"id":"12f77738f4e14e4787163a86e4f06e5143d7a74e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-08 15:16:02.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/870550 : FAILURE in 23m 31s","accounts_in_message":[],"_revision_number":3},{"id":"95817faf3399241198174ca72b93881eece93fd0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-08 15:20:06.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/50/870550/3/check/iscsi-driver-alletra-6k/4ce1fb4/4ce1fb4.html : FAILURE in 26m 01s","accounts_in_message":[],"_revision_number":3},{"id":"34e21b6d9f129072dbd23e4e90edeaf405be6bde","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-08 15:25:05.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/870550 : FAILURE in 25m 58s","accounts_in_message":[],"_revision_number":3},{"id":"181e9643ab9a276b29559b6ea052cb4074dd6e44","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-08 15:45:44.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/50/870550/3/ : FAILURE in 59m 21s","accounts_in_message":[],"_revision_number":3},{"id":"f290b8851d611d68290ef5503dbd6b99b012dcbb","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-08 16:49:28.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5d08e18d8cc94115a71e7b5ba7f938e1 : SUCCESS in 1h 37m 50s","accounts_in_message":[],"_revision_number":3},{"id":"241a469cd978e60a5c57682f0169244eb23d8750","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-08 16:51:35.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-50-870550-3/ns5_nfs : FAILURE in 1h 47m 11s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-50-870550-3/ns5_iscsi : FAILURE in 1h 46m 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"a100234b1dbd9e17f74cfdb03bdbcb9e140aa20c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 16:54:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/caa6f8301ec040f79342ce021e8076b4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/593c1cee12384d00a78f816fbdbd3e58 : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1130a6e39f634477880b6d2479a03bea : SUCCESS in 10m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/881b62e75b48437eabda891b3378b209 : SUCCESS in 10m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8bb45c6b2414a9294b36614a2731679 : FAILURE in 6m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/e5fc79b6674645898181f707e764666e : SUCCESS in 53m 57s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d44a698228d34868aece8c6da40e5719 : SUCCESS in 59m 33s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d6dcbc23867b4f5ab6e1855f979b7c86 : SUCCESS in 1h 45m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94ce58ebd16e4af584f307bba458d7a5 : SUCCESS in 1h 09m 58s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/828c46db0c254e218981c7c03343d496 : SUCCESS in 15m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c3ef27c6b6574f97946f918ed7af7da9 : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/91735d244b2e462b827ea1c17adbecf7 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5e6d1be8daab4f71b78ca2ca5a195c37 : SUCCESS in 5m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/92354a556c2741039370437cc7a5f059 : SUCCESS in 6m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/09db866a2d124557a52dd877950a7783 : SUCCESS in 50m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d52421be6cb740b1ad24243d02a524e5 : FAILURE in 6m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fa33f6dcf017417aa2f667df74000246 : SUCCESS in 1h 17m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b33af39d159c4a8a8fcba235ea127ccd : FAILURE in 1h 42m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/32e8872fa57c48b786cbe5ca65639a98 : SUCCESS in 1h 52m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/a139471cc4f9490b9307590b5318253d : SUCCESS in 1h 17m 42s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bedd8d13a8b442a99ecb6d3972b8f395 : FAILURE in 1h 33m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/03c768d21080420ab76715fde52905f6 : SUCCESS in 1h 37m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9e4193a7ddda49ee904b9ad67c33ef9b : SUCCESS in 1h 43m 03s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3b790490abcd47b39315167582966c46 : SUCCESS in 2h 33m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ae90897ade14da9850aee5d989126fd : SUCCESS in 46m 56s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62ea7830d6a24963b6f101e48a58d396 : SUCCESS in 1h 14m 38s","accounts_in_message":[],"_revision_number":3},{"id":"3793f7b1e7d95c3a98f8f2214d48372be591da6a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-08 19:42:47.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/50/870550/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/37825de/ : SUCCESS in 1h 48m 41s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/50/870550/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/86d0ec7/ : SUCCESS in 1h 42m 56s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/50/870550/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/a8fa900/ : SUCCESS in 1h 35m 36s","accounts_in_message":[],"_revision_number":3},{"id":"18c7bb2321bca1307fe99160a272c2743716e94b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-09 03:53:35.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/1435bf1f437e48bc85f2fcd4dced4b47 : SUCCESS in 2h 03m 06s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/1a886bad77fb49ed89952b99f3b975ea : FAILURE in 1h 33m 26s","accounts_in_message":[],"_revision_number":3},{"id":"cbd9292991d15c3f59180effbb21e27503e30dd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-02-13 09:20:32.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"440b7cc4d512ddcaca226dd91ab81306a7e40d87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 09:36:34.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01d00e1aca54441baa8c25450ed81726\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ecaf78ce22884004bc545a70ec744e5e : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b8349832095d4b50bfa71a30a1e92315 : SUCCESS in 12m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"011b3ffcbdc7702167d562b44bfbe59255fb1e5c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-13 09:37:55.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"5995c9542df71e8bf3f75b63bc33bdd3a56accb3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-13 10:45:09.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/870550 : FAILURE in 23m 48s","accounts_in_message":[],"_revision_number":4},{"id":"c083c22d900ef8b414e7d1be27f904bb8564b753","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-13 10:51:45.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/50/870550/4/check/iscsi-driver-alletra-6k/49a76bc/49a76bc.html : FAILURE in 26m 10s","accounts_in_message":[],"_revision_number":4},{"id":"05e2bbf7788aed9f0e5eadd744d28bbeee67d700","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-13 10:54:02.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/870550 : FAILURE in 26m 20s","accounts_in_message":[],"_revision_number":4},{"id":"7c662fe719040da0fe9733565b2e525a5cebfc87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 12:42:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c38e19fe54b84c1b9d1594fd0a40ea6d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deb9e3a6ba08421b9837b7ae2969e6fe : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/816b3cf4c9b3442fbfb7f572dc954dc4 : SUCCESS in 10m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02f40c8c46b545e6809c63d8628fef1e : SUCCESS in 13m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc3155dbaae6477eb401fa807007840e : SUCCESS in 13m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/e5e7ccdaf3a640db955ed6c5ebc0418e : SUCCESS in 1h 03m 02s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d5056ee1c0f243f9b05f673d4f84a0a0 : SUCCESS in 1h 06m 30s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c480d0007da841a8bae52950b9af65dc : SUCCESS in 1h 28m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/903e7481267e431ebc92bf0aab6f34a7 : SUCCESS in 1h 04m 49s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f10d063e09f847d3824b3ac1308c6b69 : SUCCESS in 15m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6e965a0589aa4621af747aded9096b1c : SUCCESS in 6m 06s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bc0df98d49a4469d8035c833bf24680c : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e4bbbbd6d57545f58183c78c311c2bba : SUCCESS in 5m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e1936fac568f48fd869f6a55937ee124 : SUCCESS in 5m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6503a384464d471a95b2218ee0c7b6fe : SUCCESS in 1h 09m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/93d1943a052b4f338ee3ef3c4f1757f6 : FAILURE in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/30d340d4245f4394a1b03eb0de5b7d17 : SUCCESS in 1h 25m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/31a11e0627434b1baf69731e00e3fc34 : FAILURE in 1h 29m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/bc7373e2d95640f0b986bc0ac02c4b06 : SUCCESS in 1h 40m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/83d696b4dbb347c8b95f68b6d06a1a92 : SUCCESS in 1h 06m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1dc9b36b03a94a3aa9361c9b37c1647f : SUCCESS in 1h 41m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/086d61765c3f414494d5645abab4a07e : SUCCESS in 34m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/12d58a5854c5412ebf488814a10bd21f : SUCCESS in 1h 20m 21s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1f50bd4809ea42bc8fbced774ce91956 : TIMED_OUT in 3h 04m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f812cea5d524e9aa523620d502bf02a : SUCCESS in 44m 49s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c6b9b5ec32642ddbe34d58422848502 : SUCCESS in 1h 04m 33s","accounts_in_message":[],"_revision_number":4},{"id":"c18ecd75375ee4439e18fca6d6b12b210ebc10d1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-13 12:56:04.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2be7556754454d95ac3208ec2bfac07b : SUCCESS in 1h 21m 44s","accounts_in_message":[],"_revision_number":4},{"id":"300c4dd04b23b5e35715dd9c323618824e0931b8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-13 15:01:18.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/3cd073b954c84d8db09ea369ea28b358 : SUCCESS in 2h 04m 44s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/29d24b3cd3a645c7b983b01ad0090d97 : FAILURE in 1h 56m 47s","accounts_in_message":[],"_revision_number":4},{"id":"aedc952728e14177cb8bcbe4a561bde6f5fa9e0e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-13 18:06:22.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/50/870550/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/7326d46/ : SUCCESS in 1h 46m 43s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/50/870550/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/f866d24/ : FAILURE in 2h 09m 05s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/50/870550/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/4a9bb98/ : SUCCESS in 1h 40m 01s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/50/870550/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/b080ad3/ : SUCCESS in 2h 04m 59s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy pure-devstack-cinder-tempest-fc-aio-jammy : NODE_FAILURE in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy pure-devstack-cinder-tempest-nvme-roce-aio-jammy : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c1366f7e2cd6f4d9d2bd1a252dfb9133605b9e1d","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-13 18:15:51.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/50/870550/4/ : FAILURE in 55m 13s","accounts_in_message":[],"_revision_number":4},{"id":"bf1ced947095b60fdafbd795ef7e313edf019e93","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-13 20:26:56.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/9b87f11d597f4cc4ac5802b4d0a7a887 : POST_FAILURE in 26m 50s","accounts_in_message":[],"_revision_number":4},{"id":"632ca1a3a5299c6c7127a72a9a58429e4376c81d","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-02-13 23:53:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"beaebc49a3853c53a2865c7ae3b3fe261c611fac","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-14 00:21:36.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/50/870550/4/check/DellEMC_Unity_FC/a6b1746/DellEMC_Unity_FC/2261 : SUCCESS in 1h 28m 32s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/50/870550/4/check/DellEMC_Unity_iSCSI/7828d44/DellEMC_Unity_iSCSI/2121 : SUCCESS in 1h 42m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"46c2ed845a509e638982e44b546333a822ca4a3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 00:24:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/791425506bbe415899b9bd5ab3186342\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/45fb462f3fff44e1b6d14fc4fecc9759 : SUCCESS in 29m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dc75a18b097e4e1fb1b87f7d8cb3e6dd : SUCCESS in 9m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e88a2cd8b24280e3ec9711febd885aa69565b1e8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-14 01:13:38.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5d965532ad514cdba6f2cb00d3763029 : SUCCESS in 1h 18m 25s","accounts_in_message":[],"_revision_number":4},{"id":"6f7963f3bde85a2019d3bf5626fb3818958b047e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-14 01:29:42.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-50-870550-4/ns5_nfs : FAILURE in 13h 39m 36s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-50-870550-4/ns5_iscsi : FAILURE in 13h 48m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"c4b8db85d618e5c22964050bd312a7ba2e01d38c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 02:07:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8a0bd86d3b445d986104df63d78621f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/653e79bd8a6149dabf28f6c2eceb9902 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/010e773b5c384e0194b5f2cdfa8b9d80 : SUCCESS in 10m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f21c25ef3494acc91eeace6641d178b : SUCCESS in 9m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/087f3fc8a7c54b30b85434946188b317 : SUCCESS in 12m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/ae64546ebbfb42858f09013c32efeab9 : SUCCESS in 1h 10m 34s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bffaa7dd34ef4d3d8c96b723327a08df : SUCCESS in 57m 25s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8dbc1fc5bb2c4fec99348b7b7521f746 : SUCCESS in 1h 05m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24921f8dcadb44b4b57e9977b1f26be4 : SUCCESS in 1h 03m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3a5a1f1179e44a639db578404ca6f62a : SUCCESS in 14m 57s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/414a7ef6b82b41769673c60c1f06125c : SUCCESS in 5m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f32a357f10da48f6907f49837861600d : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ff090d093e1948f480853dec13ef9a4e : SUCCESS in 5m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c83a43d0e605453fb9fea6fef5c97f85 : SUCCESS in 6m 40s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/394fdd5a461d4a7fad138e2e0f40e868 : SUCCESS in 51m 19s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/75847d887f544790ac118b3e7364407f : FAILURE in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c8b4fe53edca4244b220dc36c6b09361 : SUCCESS in 1h 03m 30s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1df36af9839d4904bae0b05307bb4dc3 : FAILURE in 1h 20m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/707ddb03bdd14d8c8e2b33a3d166f9e8 : SUCCESS in 1h 41m 58s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9c39962bf4a24859a98611ad4d07c8c0 : SUCCESS in 56m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3005b5813a734a71ac901625d3bfdd72 : SUCCESS in 1h 23m 29s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/821980b5c5164d54850ce082652195f5 : SUCCESS in 48m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b34920316164459680aaa13926120cfc : SUCCESS in 1h 29m 49s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7be273fd15d54abbbf5b13c776811aef : SUCCESS in 2h 07m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7fabf2843e84b63a66c58b19bd4ca25 : SUCCESS in 55m 59s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da33922906bd4228b8bc471a97ff0222 : SUCCESS in 1h 35m 12s","accounts_in_message":[],"_revision_number":4},{"id":"dab4b3573c86618ec6f5d7f71041ce1adffd9bf4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-14 02:54:28.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/4/upstream-check/cinder-tempest-plugin-ontap-fcp/ae555f9/ : RETRY_LIMIT in 8m 40s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/7bd44ad/ : SUCCESS in 46m 22s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/4/upstream-check/cinder-tempest-plugin-ontap-nfs/19cb49c/ : SUCCESS in 44m 37s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/7f9bf58/ : SUCCESS in 43m 58s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"1f3ed84a34e163efa071888310199d3b91cc1861","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-02-14 03:50:15.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/50/870550/4/check/DellEMC_PowerFlex/88df34e/DellEMC_PowerFlex/1602 : SUCCESS in 1h 42m 38s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"10793828ca41fa42d5de8cba3a4d742c85a94177","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-14 04:25:42.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/50/870550/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/1ab08b5/ : SUCCESS in 1h 34m 00s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/50/870550/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/6a62aa7/ : SUCCESS in 2h 01m 02s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/50/870550/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/b93020e/ : FAILURE in 2h 01m 30s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/50/870550/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/b8d7cf1/ : SUCCESS in 1h 35m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a810bcafcd595df76eb112c2f4df3393c0e7826c","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-14 12:32:08.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/0a98bb621bd54e0885889a341dca95c1 : POST_FAILURE in 32m 05s","accounts_in_message":[],"_revision_number":4},{"id":"c473b7eaf9ceb0eaf9b5501284f96a8d67e6a576","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-14 19:25:27.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/50/870550/4/check/DellEMC_Unity_FC/0c8ad00/DellEMC_Unity_FC/2270 : SUCCESS in 1h 18m 36s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/50/870550/4/check/DellEMC_Unity_iSCSI/7311367/DellEMC_Unity_iSCSI/2130 : SUCCESS in 1h 43m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"504ce7e9595111dccb3bc6b39b3feda3ab8ceef5","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-02-15 04:29:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/870550-4/HPMSA-iSCSI/ : SUCCESS in 1:35:46\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/870550-4/HPMSA-FC/ : SUCCESS in 1:17:02","accounts_in_message":[],"_revision_number":4},{"id":"250de0d8e85ce0a63c184712f283087cdb35d1b7","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-02-15 10:36:48.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/870550-4/Seagate-iSCSI/ : SUCCESS in 1:57:06\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/870550-4/Seagate-FC/ : SUCCESS in 1:35:39","accounts_in_message":[],"_revision_number":4},{"id":"17a18de79a63eb97b8808c25a90f62bd34bed909","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-02-15 10:44:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/870550-4/PVME-iSCSI/ : SUCCESS in 2:00:08\n- PVME-FC https://os-logs.seagate.com/dell-ci/870550-4/PVME-FC/ : SUCCESS in 1:37:27","accounts_in_message":[],"_revision_number":4},{"id":"0f3a3879f11141440d7cc7f37d5ad1bba6f2b9d1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-16 01:48:05.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/d8ce262dbcdc4a50ac22c318f5eca468 : SUCCESS in 1h 56m 37s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/cf973dab28214da9bf4c22c18265c28c : FAILURE in 1h 53m 04s","accounts_in_message":[],"_revision_number":4},{"id":"3ef45e7acf4d898d8d2268610cc1706b30715a78","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-02-17 06:24:48.000000000","message":"Patch Set 4:\n\nBuild succeeded \u003chttp://173.39.89.136/logs/00/870550/4/check/dsvm-tempest-cisco-zonemanager-job/0892341/\u003e","accounts_in_message":[],"_revision_number":4},{"id":"1d6021f517b3cfb0cf14cf735df52534110d396f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-02-19 03:04:34.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/4/20230219094211/NEC-V-ISCSI : SUCCESS in 2h 14m 56s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/4/20230219094211/NEC-V-FC : SUCCESS in 2h 22m 19s","accounts_in_message":[],"_revision_number":4},{"id":"9d600b085296aabba7914ea202f81e9dce89ac30","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-02-23 20:34:24.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/4/20230223223222/NEC-FC : SUCCESS in 3h 31m 42s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/4/20230223223222/NEC-ISCSI : SUCCESS in 7h 01m 58s","accounts_in_message":[],"_revision_number":4},{"id":"1d4c3113ace579f7e5a1dba190c45a6e2bd4d680","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-03-01 12:49:08.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"05733a579476f578e4ec42dfa60f06a1ca5169c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 13:03:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4116e2530aa46f9b4ff41d75dc604ae\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/23ad0a6a13584ca59cba9c42947de4c5 : SUCCESS in 10m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eafb1719349d42f0b57a420b53679713 : SUCCESS in 12m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"62c5294db5abe6cb074aeb915eb0ed0c0af85453","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-01 14:13:46.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/870550 : FAILURE in 23m 49s","accounts_in_message":[],"_revision_number":5},{"id":"e15073e2c9d43c381c82aad136d0d73614b7b528","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-01 14:33:03.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/50/870550/5/check/iscsi-driver-alletra-6k/b016e42/b016e42.html : FAILURE in 26m 10s","accounts_in_message":[],"_revision_number":5},{"id":"bbde33ad2e67c305d5106017f29c6adb465b6afd","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-01 14:38:12.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/870550 : FAILURE in 26m 04s","accounts_in_message":[],"_revision_number":5},{"id":"f713eea408d2a2bf6ce71f04af6c1b69bc76ebf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 15:40:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf6258f1c4314dfa8f04241c33d5caa3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/674a7721e54543c3b15f12cad3c04e4f : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56bca0ddabcf466d9a7b891f3d6a95e1 : SUCCESS in 10m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dda4441aa64a45b39f193377262a86b3 : SUCCESS in 10m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cd7249a64fa48a4baf842eb808a8da2 : SUCCESS in 12m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/24daac67ade643fd99062de309e91c1f : SUCCESS in 58m 54s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bfb46f3a34054540a0983fbcb20b907c : SUCCESS in 57m 27s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/909776d66bba4658876cb5cb3b3aa06c : SUCCESS in 1h 10m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c875cb7b695542f28a9272d72e218367 : SUCCESS in 59m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dd6b220891dd43ff811d8fa51a994f9f : SUCCESS in 15m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/93496319a5c7495686583d5827b98f81 : SUCCESS in 5m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6435cf8e7bb145b1bdd59e7fe05b6203 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0e51742e08b146e9aa296f0bdc79503d : SUCCESS in 5m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3858b0e279cf4a0ab0ed938c384ad151 : SUCCESS in 6m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ad38892dbfa747ef8138ad26552bc14e : SUCCESS in 52m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/267bcffad7924064990b7eaf4d080c10 : FAILURE in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8dc4c7612444445584f890408f5373d7 : FAILURE in 1h 20m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dcb73458f99c4185b5e8a01ed57aa9b9 : FAILURE in 1h 33m 22s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/98649cc9711f41af8137e850313efdc1 : SUCCESS in 1h 38m 05s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c4aee87bfd344c84a59d7334f5fe50dd : SUCCESS in 1h 02m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5de21b3847f048d5b64d9c57d505727c : SUCCESS in 1h 54m 34s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5d90d15faad24608849758d0788a0615 : SUCCESS in 51m 34s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0ed0c130c4a347bc8c28b93c56f15411 : SUCCESS in 1h 11m 42s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5a4b13485cee43bd90dd429901034961 : SUCCESS in 2h 42m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1bdafd1188264d7ea21e29cfdd514f43 : SUCCESS in 1h 11m 53s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a55149a6f7204bbca99b5bfa1d31983c : SUCCESS in 1h 32m 34s","accounts_in_message":[],"_revision_number":5},{"id":"a0e309e600e1b641747fd1b6efc0a6e391e76230","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-01 16:22:19.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/75dcdadf404a456084eaa6528ac95ce3 : DISK_FULL in 1h 15m 53s","accounts_in_message":[],"_revision_number":5},{"id":"821a05b4883b797ccf20840add07d945751cd2eb","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-01 19:48:25.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-50-870550-5/ns5_nfs : FAILURE in 4h 11m 54s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-50-870550-5/ns5_iscsi : FAILURE in 5h 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"576ad02f50dd2a2b02622b14e29396cc19445387","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-01 22:22:18.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"259a7322590f498880d2802dc8e1e0be8b08f95b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-01 22:30:02.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/50/870550/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/ade3478/ : SUCCESS in 1h 39m 54s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/50/870550/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/cf9f568/ : FAILURE in 2h 11m 22s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/50/870550/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/4d38267/ : SUCCESS in 2h 03m 52s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/50/870550/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/2bc4401/ : SUCCESS in 1h 41m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d38e364f932c878f3c3cf01693ca0e9d5fc8eabb","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-03-01 23:19:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"12945e124b79c6123868bdab9aae77b6eb023b0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 23:40:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57674a8b4d1e4abaa5288fa9fc05f460\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd921546d5914d4bb398b79723653cb2 : SUCCESS in 18m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3ac247e1ddff4ed49eb3bfd9db0a33e5 : SUCCESS in 18m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"80ea8ca60f4ef32dfeecd4c1b5d59ca418793ab7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-01 23:48:55.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"8e31cd9b0c7eb874b131e2df10e944c506168a5d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-02 00:33:21.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fc9745f7868f4e91be94cc7422029ff1 : SUCCESS in 1h 12m 17s","accounts_in_message":[],"_revision_number":5},{"id":"170f1d0faec45420bd3dcb463dd0455bdb844e42","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-03-02 01:34:07.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/50/870550/5/ : FAILURE in 56m 02s","accounts_in_message":[],"_revision_number":5},{"id":"acb60193cafd864a348bc3bddbbe5f3b867a08bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 01:45:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/308b78171afd40ffb747fe09000e415d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/322bbd38cf984c28b2c125593cac8569 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7e7403e48c949008094ad8155029550 : SUCCESS in 14m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79887d2da42148e7ab7aaa42d8f499ef : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c015d85f594a4642a76fc4c38aeea6f8 : SUCCESS in 10m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/f5b6ae1fe6ec4c0ea4f6841ddbe87809 : SUCCESS in 54m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c9f7568f3daf4eb2912e271c357c7cc1 : SUCCESS in 50m 38s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a5fce4d3c75c40e08edfd3d65b015843 : SUCCESS in 1h 29m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/345a3f701e994cebbc631ca3317ff204 : SUCCESS in 53m 47s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e218baa2f1b740b2a5f9062910500dc3 : SUCCESS in 16m 13s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/02953d92c2fd45e0bd0febd9af8ecef2 : SUCCESS in 5m 07s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f443587f95cc4b968afca865cd36804e : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fd69b3dc896b44fc8c89b9f7a2ddbdb5 : SUCCESS in 6m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c9b27f2993fa47bdacfc6266f07700d1 : SUCCESS in 5m 13s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f9a2211ffd394c47a48c4059ff06a905 : SUCCESS in 40m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9a6eca0b887744e090506dafb88259af : FAILURE in 5m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9d029a9d7ace46a6a231fec772f4561d : SUCCESS in 1h 22m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/21599c50463b4c3f9adf5847ada1f8e9 : FAILURE in 1h 24m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/85527a51c03342a187317793e268e667 : SUCCESS in 1h 47m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/dffdbef00a944e4caa38c985b271a2a7 : SUCCESS in 1h 17m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d9c8723f05094fa6acc79c90eeaaa0c6 : SUCCESS in 1h 31m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fef2f0c2c29b4ff6af4d3742c7b25907 : SUCCESS in 41m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0f08635371874a278856296225db8f16 : SUCCESS in 1h 19m 07s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a4fd9ee5a9ba4fac87981192e4301a5f : SUCCESS in 2h 20m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7333bec63cd4728b07bc561f1a52104 : SUCCESS in 52m 06s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f6468b7ce164e5f8bb2056a27dc37ef : SUCCESS in 1h 24m 07s","accounts_in_message":[],"_revision_number":5},{"id":"4f9f000b2f8330998ab91076c462cc40f9353419","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-03-02 01:53:42.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/02_03_2023_01_45_56\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"9154db2aba409c97c154a6c0fd4ad3bad44eadcd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-03-02 03:09:52.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/5/upstream-check/cinder-tempest-plugin-ontap-fcp/ca60172/ : SUCCESS in 1h 14m 29s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/5/upstream-check/cinder-tempest-plugin-ontap-iscsi/f1e7135/ : SUCCESS in 1h 17m 27s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/5/upstream-check/cinder-tempest-plugin-ontap-nfs/37c2e97/ : SUCCESS in 1h 17m 01s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/5/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/ac69119/ : FAILURE in 1h 16m 00s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"c8d204566ad8f018936484617e4a27ce3b397823","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-02 08:47:43.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/50/870550/5/check/DellEMC_Unity_FC/85aced1/DellEMC_Unity_FC/2397 : SUCCESS in 1h 36m 46s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/50/870550/5/check/DellEMC_Unity_iSCSI/ed3a18d/DellEMC_Unity_iSCSI/2249 : SUCCESS in 1h 55m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"72794819365692ca9373a6425348ebf9cbac1897","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-02 12:33:56.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/50/870550/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a3b4527/ : SUCCESS in 1h 34m 34s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/50/870550/5/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/9341246/ : FAILURE in 1h 55m 30s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/50/870550/5/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/9130625/ : SUCCESS in 2h 04m 50s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/50/870550/5/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/f7b6171/ : SUCCESS in 1h 45m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4aeb9f6fc4ccef6d89bd58de8543fa7081a4233a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-03-02 12:36:51.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"78335d97dde14d781dc2adc3abc3b31c3898902b","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-03-02 12:38:39.000000000","message":"Patch Set 5: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5ca3da646c03c46a4c0a98220e114767ee71175f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-03-02 16:28:51.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/50/870550/5/check/DellEMC_PowerFlex/b35661c/DellEMC_PowerFlex/1686 : SUCCESS in 1h 30m 55s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"456d36edeb244fdbf95adc258f2e1257da6289f1","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-03-02 16:49:17.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/50/870550/5/check/DellEMC_PowerStore_iSCSI/145781c/DellEMC_PowerStore_iSCSI/1080 : SUCCESS in 1h 13m 36s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/50/870550/5/check/DellEMC_PowerStore_FC/ee93fb0/DellEMC_PowerStore_FC/1203 : SUCCESS in 1h 33m 15s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/50/870550/5/check/DellEMC_PowerStore_NFS/5b6945e/DellEMC_PowerStore_NFS/719 : SUCCESS in 1h 18m 45s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/50/870550/5/check/DellEMC_PowerStore_NVMeTCP/7e34ce4/DellEMC_PowerStore_NVMeTCP/584 : SUCCESS in 1h 26m 27s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"acd0a7532fc5e56265f3b33f850886a1bbe0b64d","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2023-03-02 18:07:09.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/50/870550/5/check/DellEMC_SC_FC/f2b2a5b/DellEMC_SC_FC/1151 : SUCCESS in 1h 08m 38s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/50/870550/5/check/DellEMC_SC_iSCSI/5811a9c/DellEMC_SC_iSCSI/1135 : SUCCESS in 1h 28m 44s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":5},{"id":"13c269a1ed76a49ec28d9ceda546428a1f5e164e","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-03-04 18:05:41.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/870550-5/HPMSA-iSCSI/ : SUCCESS in 1:37:04\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/870550-5/HPMSA-FC/ : SUCCESS in 1:16:56","accounts_in_message":[],"_revision_number":5},{"id":"16040d91fa9e711f4ebe31e5ebe52b4c22f6d6da","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-03-05 02:22:45.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/870550-5/PVME-iSCSI/ : SUCCESS in 1:50:06\n- PVME-FC https://os-logs.seagate.com/dell-ci/870550-5/PVME-FC/ : SUCCESS in 1:35:19","accounts_in_message":[],"_revision_number":5},{"id":"e07ea1419669885b4f4800fbe7d833db5fd21f03","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-03-05 06:28:33.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/870550-5/Seagate-iSCSI/ : SUCCESS in 1:49:40\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/870550-5/Seagate-FC/ : SUCCESS in 1:36:08","accounts_in_message":[],"_revision_number":5},{"id":"9ef064a4f22a854f37e383eb62c4f94dc0170c4a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-03-06 06:10:36.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/5/20230306121400/NEC-V-ISCSI : SUCCESS in 2h 32m 28s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/5/20230306121400/NEC-V-FC : SUCCESS in 2h 56m 32s","accounts_in_message":[],"_revision_number":5},{"id":"530b557bd862774dbab3e52c50e29a156ff6ceba","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-03-07 22:36:52.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/5/20230307235609/NEC-FC : SUCCESS in 2h 34m 20s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/5/20230307235609/NEC-ISCSI : SUCCESS in 7h 40m 40s","accounts_in_message":[],"_revision_number":5},{"id":"030d47a59780ecdf93d1a72b5419e492658da426","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2023-03-08 22:09:27.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7111a8e6c20972cb6541344cd80b88a2649d999d","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-03-09 23:57:54.000000000","message":"Patch Set 5:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/50/870550/5/fc : FAILURE in 1h 07m 33s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/50/870550/5/iscsi : FAILURE in 1h 06m 04s","accounts_in_message":[],"_revision_number":5},{"id":"df59fea815791236ae1486004a8f18852bbac133","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-23 08:53:42.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/50/870550/5/check/iscsi-driver-nimble/62fac63/62fac63.html : FAILURE in 39m 01s","accounts_in_message":[],"_revision_number":5},{"id":"892bd5f8a4a882e6450cac9a88dd19bf3aa2346c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-05-08 19:32:02.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"715a7adb1880216d86c6847526d986eae0be38db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 19:59:59.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f6ede6ed671462680bcc9cc3f849190\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dadeb0eeecd44d8b85ff5154045cefe5 : SUCCESS in 25m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/265ab888c67a4d0688cf68d82bfee7aa : SUCCESS in 25m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"596da5a237843b8ee491997af0f224884075ea4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 21:52:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0668335699b47e4bf2c1fe0880b5f1f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e55eb55f2ed940789685c5cf1770b32d : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32ddb942174b431cac59ed6c0fe9e45c : SUCCESS in 12m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ffa398eb4ac04a579bfbc8d90c16298c : SUCCESS in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b19a28ea0ab84d0fab8b57c094f08a51 : SUCCESS in 11m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/21bbd7dbf6784d6b9acbbf1b6be3f626 : SUCCESS in 55m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bbb86a177f63480c93986c2bb11846ba : SUCCESS in 1h 27m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4052468c00a5409da02657727c569e8f : SUCCESS in 59m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4c1e0f943b77400b8a569994000e98fd : SUCCESS in 18m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f838256d9889488baf9fcb10ce7a57b9 : SUCCESS in 4m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f5ef684e981a48108285a77d94de9f34 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/af7d2b3fa1b6485faaeb6e326712c0a3 : SUCCESS in 5m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9e3b41a5b0f14aaf80cd4332ea9777b3 : SUCCESS in 6m 15s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ccb2ba1d7f904457846dd452141f1931 : SUCCESS in 54m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/02367aaa3dad4006b76658a4fbb492ca : SUCCESS in 7m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/029abac465e64a92aaab027550665517 : SUCCESS in 1h 59m 47s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9253fa78a147498c9cd894a131823160 : FAILURE in 1h 46m 21s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e5866313e2e044a2ba7092b846e2dfbd : SUCCESS in 1h 51m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/05be7163d60c427fa9d1843740244bba : SUCCESS in 1h 57m 54s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/711f73dc3d4b463bb4c3b9d68f36857f : SUCCESS in 1h 06m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/13132ba77b9f4f7b990da313df7ffabf : SUCCESS in 1h 30m 20s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/48a4e12e89ec467282848e7fdcdc7970 : SUCCESS in 47m 26s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/68fe9be95a4c42609e27ef92600cbc73 : SUCCESS in 1h 20m 03s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8c82cb99a25046608c85f36ee7bfdb98 : SUCCESS in 1h 44m 40s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4a563de31db04649991ccfc33742e0ca : SUCCESS in 2h 14m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8adbcb934764c1198a121a075c384e4 : SUCCESS in 54m 05s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54f99b492ad4456c886a3c59adb2a8fc : SUCCESS in 1h 34m 45s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/521f2d05b7f042b496e392eecca0963f : SUCCESS in 11m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8a8fa2e9944d4a0bef63e9c78c81cc50d1088439","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-05-08 21:54:32.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"3e2f9625443d8251ceb0c11f337f14f6287a3336","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-05-08 22:16:51.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"dfd9c8fe2323771087553d2ecbaff5994b55965a","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-05-08 22:17:21.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"d3df0d378a84694cbd28ab5c8cbcf3d3836b2d83","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-08 23:16:15.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/41a281126e954e6b9aa22c4385f1826d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e597c6ba282f4a1ea3082576c77125ba : FAILURE in 1h 07m 14s","accounts_in_message":[],"_revision_number":6},{"id":"b84e2b0c3ab4836989c29f6b88310134335893a4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-08 23:20:27.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/6/upstream-check/cinder-tempest-plugin-ontap-fcp/8bd230f/ : RETRY_LIMIT in 13m 22s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/eef50e4/ : SUCCESS in 46m 20s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/6/upstream-check/cinder-tempest-plugin-ontap-nfs/475ceda/ : SUCCESS in 1h 26m 33s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/6/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/29ae803/ : FAILURE in 1h 25m 10s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"596ebdc6a2f18a782a660dbd5d2f19f6f2f16343","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-05-08 23:34:51.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/50/870550/6/check/DellEMC_PowerFlex/f0e30bd/DellEMC_PowerFlex/67 : FAILURE in 1h 41m 35s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"8d212cea68ade4aac98917fd37a1c60da24f1940","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-05-08 23:41:09.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/870550 : FAILURE in 59m 33s","accounts_in_message":[],"_revision_number":6},{"id":"8272331c4cfd905cba18f71b527e5b98014322d7","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-05-09 00:05:00.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/50/870550/6/check/DellEMC_PowerStore_iSCSI/3cbe12e/DellEMC_PowerStore_iSCSI/24 : SUCCESS in 48m 16s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/50/870550/6/check/DellEMC_PowerStore_FC/105661e/DellEMC_PowerStore_FC/25 : FAILURE in 1h 07m 26s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/50/870550/6/check/DellEMC_PowerStore_NFS/b4d34ab/DellEMC_PowerStore_NFS/23 : SUCCESS in 53m 41s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/50/870550/6/check/DellEMC_PowerStore_NVMeTCP/c85f68a/DellEMC_PowerStore_NVMeTCP/23 : SUCCESS in 1h 03m 24s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"bc681b605fbb53c0d72d9e6d4f9b0d9c398a2a47","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-09 00:44:42.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/50/870550/6/check/iscsi-driver-alletra-6k/b024bc2/b024bc2.html : FAILURE in 2h 01m 24s","accounts_in_message":[],"_revision_number":6},{"id":"63642b1afd3c21dd060d5f5c6943c0ff80130c7c","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-09 01:02:30.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/77be6703b3534add8b3879ce8473a7f0 : POST_FAILURE in 50m 34s","accounts_in_message":[],"_revision_number":6},{"id":"85b0dbf60150f4c40ba727b27074d84a93ea8f1c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-05-09 01:12:29.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-50-870550-6/FC/ : SUCCESS in 4h 25m 4s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-50-870550-6/iSCSI/ : SUCCESS in 3h 56m 38s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"728175f3ccebed5d1c268ff6f4168ce8a5f01acf","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-05-09 01:18:14.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-50-870550-6/ns5_nfs : SUCCESS in 3h 48m 31s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-50-870550-6/ns5_iscsi : SUCCESS in 3h 52m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"962566b8782dffcaf6bd68936263bdfc1116a51a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-09 08:08:14.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/50/870550/6/check/DellEMC_Unity_FC/47aff1a/DellEMC_Unity_FC/110 : SUCCESS in 1h 39m 31s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/50/870550/6/check/DellEMC_Unity_iSCSI/c895eb6/DellEMC_Unity_iSCSI/113 : SUCCESS in 2h 06m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"b2e23a22e89f0ce1f5fa9726a60f573b2cb5081e","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-05-09 18:54:54.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/50/870550/6/check/DellEMC_PowerMAX_FC/a0f4c87/DellEMC_PowerMAX_FC/51 : SUCCESS in 5h 29m 53s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/50/870550/6/check/DellEMC_PowerMAX_iSCSI/d3e7ac1/DellEMC_PowerMAX_iSCSI/54 : ABORTED in 10h 30m 05s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"48448be024b88f448b3dda2f8d1ac760656d8097","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-05-10 02:16:28.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/6/20230510052803/NEC-V-ISCSI : SUCCESS in 2h 24m 43s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/6/20230510052803/NEC-V-FC : SUCCESS in 5h 48m 22s","accounts_in_message":[],"_revision_number":6},{"id":"cc431579736e6187899cd8f912540e59bc6867c9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-05-10 06:22:53.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/50/870550/6/check/tempest-dsvm-full-focal-py3/b36fb79/ : FAILURE in 1h 07m 24s\n- devstack-plugin-ceph-dsvm-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/50/870550/6/check/devstack-plugin-ceph-dsvm-focal-py3/8950478/ : FAILURE in 44m 13s","accounts_in_message":[],"_revision_number":6},{"id":"6a86be8b1149dbf92885e2dc9168357e6f580406","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-10 09:27:34.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"628f1c8485f0a0f1add9393a347251d45193fba3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-05-10 17:09:44.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/6/20230510190032/NEC-ISCSI : SUCCESS in 3h 11m 02s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/6/20230510190032/NEC-FC : SUCCESS in 7h 09m 09s","accounts_in_message":[],"_revision_number":6},{"id":"6d0ec3380ba1ecb710e6584dfa2b656e3c57a0fd","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-05-13 13:57:55.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/50/870550/6/ : RETRY_LIMIT in 3m 05s","accounts_in_message":[],"_revision_number":6},{"id":"30791c643e533115028e303c3a8aad3d10aec58e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-12-06 06:26:30.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"9b8bd8cca3b5008b33f65362e0b2bc6932757f1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 06:54:40.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/423c345e500e4782820e5db16bbb239b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c639e21badfa4a2984b2192e170525d2 : SUCCESS in 25m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d96e6d3f29714cdbad399d117026e9b7 : SUCCESS in 10m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/26099717300749d9afb95dcbb71b6ea4 : SUCCESS in 23m 16s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a347d3ee126ed517118d6fc1960f36d33517b42c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-12-06 08:45:10.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/ca5368c7ecd346a9adf3c9851d9f9f0b : RETRY_LIMIT in 52s\n- pure-devstack-cinder-tempest-fc-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/6dd33b92b6644876b32ab5cf077c66f0 : RETRY_LIMIT in 51s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/12131b76f82e44709fa79b43f7b388dc : RETRY_LIMIT in 52s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/89554c012537460181420dc2e2ef75bc : RETRY_LIMIT in 51s","accounts_in_message":[],"_revision_number":7},{"id":"804cbafce77ee427c6d8c833d9a1b1379aea9f0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 09:03:49.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9cb137cc94f492384027bab29c7b5f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/993fa494b12b4dfaa7e04df602b96733 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1286d35aaa4408f8d426934ed31d5ed : SUCCESS in 12m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a25416342ac04180952c9f6479d8634e : SUCCESS in 9m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dbea1aa7118413aa940802fe3364775 : SUCCESS in 13m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/a8e90adedeab4edabd78cc00488b0afb : SUCCESS in 59m 08s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/25a421cb44cb45d59d24f62d779b0ad8 : SUCCESS in 47m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/64468e3323c34c89b47a7e28936f8e5a : SUCCESS in 1h 40m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68669a06bf5b44ffb1425a3dd1c2ac7f : SUCCESS in 1h 09m 44s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d36e34de0d6e4de8bd96518461047e37 : SUCCESS in 17m 12s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e893a163a38d4237b5d5d2f2e4634609 : SUCCESS in 14m 43s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a585740c6bcf4d8590de9bd9da8e45d1 : SUCCESS in 5m 52s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/72efa9ea240e4371a166736ed98d832c : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3105e9ba40074ceab766f8eb3098b2e2 : SUCCESS in 7m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f4a77dfacc4246f582a6e2ebb130ce02 : SUCCESS in 7m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4646333c5d93497fb915466e88dce367 : SUCCESS in 42m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/200161c91d6a4cc4b2cc686326170060 : SUCCESS in 7m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ab9ff7bf41d3429da452f77358629639 : SUCCESS in 1h 51m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7e0e359ef59b470aa8ab92267eb53549 : FAILURE in 2h 27m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ad8b655add1344af9b50188477c203ef : SUCCESS in 1h 28m 23s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/0f9b6694732f47c3aca1b561b741d61c : SUCCESS in 1h 43m 36s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ed01cc437ac441ed86dca5d3bccce6de : SUCCESS in 1h 12m 13s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dbea3b1fd6114baaa765dbc6b85a206f : FAILURE in 1h 27m 19s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4397cf03e1674d21b16afc5d7c001ae0 : SUCCESS in 31m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5a5ff002323746a3be391acbb1fc6317 : SUCCESS in 1h 23m 44s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/7f8b39f69ee74f1598ababab1817a883 : SUCCESS in 1h 33m 02s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c8e35c5d9b1b452aa334b2e076370f97 : SUCCESS in 1h 14m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67c58b0721eb47deb1f1c8d1aabc5119 : SUCCESS in 48m 32s","accounts_in_message":[],"_revision_number":7},{"id":"9ffa1d7a652048965bf42e46251ac100c108e596","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-12-06 09:36:12.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/4450f73d3ac3456db4669682c2be9193\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e530a0168b79403886e475de9db29882 : FAILURE in 1h 26m 21s","accounts_in_message":[],"_revision_number":7},{"id":"b3d225aba308def2acb6fcfe1d4b0c152ba85f94","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-12-06 10:03:03.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-50-870550-7/FC/ : FAILURE in 2h 6m 8s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-50-870550-7/iSCSI/ : FAILURE in 1h 43m 25s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"2765859848a64dd18442049841e6ac09019e26f6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-12-06 11:18:36.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/50/870550/7/check/DellEMC_PowerFlex/195196e/DellEMC_PowerFlex/995 : SUCCESS in 1h 04m 36s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"69d8ddcc22686d52e0af0cc9fdfffd8d0c868369","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-12-06 12:05:48.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/50/870550/7/check/DellEMC_PowerStore_iSCSI/19f8f51/DellEMC_PowerStore_iSCSI/448 : SUCCESS in 1h 02m 14s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/50/870550/7/check/DellEMC_PowerStore_FC/12a8a7a/DellEMC_PowerStore_FC/451 : SUCCESS in 1h 40m 04s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/50/870550/7/check/DellEMC_PowerStore_NFS/550bdb0/DellEMC_PowerStore_NFS/457 : SUCCESS in 1h 00m 17s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/50/870550/7/check/DellEMC_PowerStore_NVMeTCP/9606f86/DellEMC_PowerStore_NVMeTCP/441 : SUCCESS in 1h 33m 13s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"792cc681c86b1253b33cae15248b54d9b473c0d2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-12-06 12:17:20.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/7/upstream-check/cinder-tempest-plugin-ontap-fcp/416766b/ : SUCCESS in 3h 12m 23s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/7/upstream-check/cinder-tempest-plugin-ontap-iscsi/98e0f16/ : SUCCESS in 2h 17m 41s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/7/upstream-check/cinder-tempest-plugin-ontap-nfs/d994b4e/ : SUCCESS in 45m 22s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/7/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/3eb7a83/ : SUCCESS in 2h 21m 43s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"f551813d0f789c1e4d294c8f1e0b69554edb86dd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-06 12:59:05.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f84dd034018d4ee98ede5b0c52f57856 : RETRY_LIMIT in 5m 38s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/ec95e4c397224e1ab129b5c134d5511e : RETRY_LIMIT in 6m 05s","accounts_in_message":[],"_revision_number":7},{"id":"853558568a735e165484d13df7d4b4b854d69f73","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2023-12-06 14:28:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"97275b6a22fdf2e6bab32f382128adec34f79e48","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-06 16:26:53.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/50/870550/7/check/DellEMC_Unity_FC/59c95e4/DellEMC_Unity_FC/1049 : FAILURE in 2h 30m 50s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/50/870550/7/check/DellEMC_Unity_iSCSI/b41c9e8/DellEMC_Unity_iSCSI/1063 : FAILURE in 2h 20m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"59812ae98f919300c8b7100d6c66a9a01b5fd0aa","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-12-06 19:39:55.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/50/870550/7/ : SUCCESS in 1h 30m 14s","accounts_in_message":[],"_revision_number":7},{"id":"0849f7a9206c5d9490de0e128f1e6cc91139a4c3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-12-06 20:31:13.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f5451988cd1e497b90f6991e020332d3 : SUCCESS in 1h 19m 39s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/3118dac3df954c1984d75e1225550611 : SUCCESS in 1h 03m 49s","accounts_in_message":[],"_revision_number":7},{"id":"ded49a5081c1259c785ba12cfb039a992455d478","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-12-07 19:07:52.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/7/20231208003101/NEC-V-ISCSI : SUCCESS in 2h 56m 08s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/7/20231208003101/NEC-V-FC : SUCCESS in 3h 36m 47s","accounts_in_message":[],"_revision_number":7},{"id":"93cc523161b575820feee957c000d297f63e17ad","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-12-08 19:12:25.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/7/20231209001924/NEC-FC : SUCCESS in 3h 48m 46s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/7/20231209001924/NEC-ISCSI : SUCCESS in 3h 52m 58s","accounts_in_message":[],"_revision_number":7},{"id":"a406792dec4887433d10ffeff86e0161e765065b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"date":"2024-01-18 12:56:47.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"13ccd5ce9be14658bef86a7690094c58f9eac329","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-01-18 12:57:15.000000000","message":"Patch Set 8:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16385/","accounts_in_message":[],"_revision_number":8},{"id":"bc8449b5d4a90c886ea2ac970178c2a4e5e8b91b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-01-18 12:57:18.000000000","message":"Patch Set 8:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-cinder-zuul/16385/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"345c3753b2a3d1fcb33e7a99dc9cfb3cac5b7b51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 13:40:35.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/218ea9ff365a4a8daa07e21cf9b2c125\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/777a79ccab7d4378809e6684fac1fc8a : FAILURE in 39m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0d441e018e7c49f3988af611b155b6a9 : FAILURE in 40m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0704eeeef58549bfa7d569ff403cec0d : SUCCESS in 9m 22s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0812b400cd5ec085d01eb7c98884fcc75b1d483d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-01-18 14:43:33.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/d0317157c65a416e94e612842968db62\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a534dc3b656c4cda91a1992342622865 : FAILURE in 1h 26m 52s","accounts_in_message":[],"_revision_number":8},{"id":"00d829a227e769f72731a96026ce48462e83e684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 15:18:11.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03caae7c46414afda68ecf85e3fbdd59\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3583322355a46a0876dfb9873ff490a : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7cc2ca9028043c29d5fcb0d264786f0 : SUCCESS in 11m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0cc1a31a130c472494599f1c46cbd079 : SUCCESS in 10m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b4d1ccf085c4a7bbd586316ab316bac : SUCCESS in 13m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/661e2eac2c0e4721b79874d15c4b6145 : SUCCESS in 1h 04m 14s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/17dff31b3a0648179e64156a5422bfbc : SUCCESS in 1h 10m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3990e59e72944022ac52b256e34a5789 : SUCCESS in 1h 42m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7bf1384610f4d92acdff5f36dc27191 : SUCCESS in 25m 06s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7d52d2f540054c1aa4659329ee7526a2 : SUCCESS in 17m 09s (non-voting)\n- cinder-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c7a794207878467fbd08b078ec0ba5aa : SUCCESS in 13m 42s\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/0daf1aafa9ce49c4b9aa13ac49a7a225 : SUCCESS in 5m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c715e9513a3c4bf98aa24ed3de9396dc : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eb72e4d65f31462587018d51a10d996a : SUCCESS in 7m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/828e10b3193d46e6b8482ad94ceedd9f : SUCCESS in 7m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/02872b76975d44cb8349531a87430f18 : SUCCESS in 55m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8bf5db0383744538b5263b519ea3d3e0 : SUCCESS in 5m 34s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7f26b2d7fef54eafb038985e12be6457 : SUCCESS in 2h 12m 43s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/57746760a6f54bbcbc704c4f6a505b90 : FAILURE in 2h 04m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/52ef7d8b35eb4bfe85142cd586a38667 : SUCCESS in 1h 50m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3cd60514e0f749f78aabad9564cec32f : SUCCESS in 1h 49m 15s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c365729740db4a86ac0e0982a8dd251d : SUCCESS in 1h 15m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1f989b9759dc495db63ed3015899e89f : SUCCESS in 1h 38m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/4c054e77504f42fdad84b6ff1a3cf2f1 : SUCCESS in 40m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c82a0e19af08425aaafea45cb2ad2ff2 : SUCCESS in 1h 22m 07s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/90126afceb2b44dab7b6eb0da7d31a9f : SUCCESS in 1h 39m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/85f00af8f75d42fcbaade6693f45f1bb : SUCCESS in 1h 45m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4144ecc8c26b4b068fd72434492f541f : SUCCESS in 54m 14s","accounts_in_message":[],"_revision_number":8},{"id":"a962de728e8704a9b21c5492c75ec6264fcad6a7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-01-18 15:20:55.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/870550 : FAILURE in 42m 06s","accounts_in_message":[],"_revision_number":8},{"id":"f67df475beccd9ddfa6b9ecf067507e7626d0c0f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-01-18 16:55:11.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5015fb5c8ad242ce89437b41de88a11b : SUCCESS in 1h 10m 01s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/12e40debd5134957a6bc92d1179d1d69 : SUCCESS in 1h 10m 29s","accounts_in_message":[],"_revision_number":8},{"id":"c6d8fc68bee52083d3957446627f0b8989a3e9f6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-01-18 18:02:09.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/8/upstream-check/cinder-tempest-plugin-ontap-fcp/66ecc49/ : RETRY_LIMIT in 11m 20s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/8/upstream-check/cinder-tempest-plugin-ontap-iscsi/84299cc/ : SUCCESS in 2h 39m 14s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/8/upstream-check/cinder-tempest-plugin-ontap-nfs/37b2d9e/ : SUCCESS in 40m 13s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/8/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/702bbd4/ : SUCCESS in 41m 54s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"1a1ae46faebb020f01d5a9f588fe2ab3669f4060","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-01-18 20:41:28.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/460/.","accounts_in_message":[],"_revision_number":8},{"id":"10afd99ac349a98e0d2468046e12ff0b2b1de902","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-01-19 01:37:10.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/50/870550/8/check/DellEMC_PowerStore_iSCSI/6cb7445/DellEMC_PowerStore_iSCSI/961 : SUCCESS in 1h 03m 42s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/50/870550/8/check/DellEMC_PowerStore_FC/532c4be/DellEMC_PowerStore_FC/1117 : SUCCESS in 1h 27m 07s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/50/870550/8/check/DellEMC_PowerStore_NFS/ce9c44e/DellEMC_PowerStore_NFS/647 : SUCCESS in 59m 29s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/50/870550/8/check/DellEMC_PowerStore_NVMeTCP/b2d2e37/DellEMC_PowerStore_NVMeTCP/513 : FAILURE in 2h 07m 49s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"1667cf30e7597548dfe668b3c5b76196fa25de7c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-19 06:03:09.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/50/870550/8/check/DellEMC_PowerFlex_3.6/bee16f7/DellEMC_PowerFlex_3.6/1248 : FAILURE in 1h 16m 51s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/50/870550/8/check/DellEMC_PowerFlex_4.5/64b9bc5/DellEMC_PowerFlex_4.5/219 : SUCCESS in 1h 53m 49s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"ab2c9fb6a96cc0cbcda6262d99363cb6193da92a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-01-19 07:30:34.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/456/.","accounts_in_message":[],"_revision_number":8},{"id":"7204e3d134afa32251ec73c6fbb5f8106b6b20fe","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-01-19 11:00:07.000000000","message":"Patch Set 8:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/50/870550/8/fc : SUCCESS in 0h 41m 04s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/50/870550/8/iscsi : SUCCESS in 1h 01m 39s","accounts_in_message":[],"_revision_number":8},{"id":"274fdce4fb8c47c53a36ef6176b726faa9894974","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2024-01-19 14:12:02.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"75ff55c2fc3d28c986243cf102a98d9b86c29bc2","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-01-21 11:31:10.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/8/20240121170146/NEC-V-ISCSI : SUCCESS in 3h 12m 08s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/8/20240121170146/NEC-V-FC : SUCCESS in 3h 29m 21s","accounts_in_message":[],"_revision_number":8},{"id":"7590a76c0abbdc687727404b3ef596c1cbd345a9","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-01-22 22:48:07.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/8/20240123035418/NEC-FC : SUCCESS in 3h 53m 16s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/50/870550/8/20240123035418/NEC-ISCSI : SUCCESS in 3h 53m 46s","accounts_in_message":[],"_revision_number":8},{"id":"09949019cbbb5ddc2e8357b690bcda3324721ac4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-02 02:46:37.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"47961c793e069a9594d7722d17668fee505c78bc","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-04-22 14:41:56.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"0199ba0b6ae254d9c87f5646b1fb5ccd755139c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 14:42:52.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"dcde9587e10e873709b18e95f445a4a4a91ed1ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 16:57:21.000000000","message":"Patch Set 8: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/803823, https://review.opendev.org/787296 are needed.","accounts_in_message":[],"_revision_number":8},{"id":"da3e032936cf2f2897bb0131ea8115dd592684c3","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-05-28 11:46:43.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fd256ab7519de4a83ff0b996b506fdfc244ea984","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-28 11:48:10.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"03d8fcc7453b198fa7a33b3731d95801a974b923","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2024-05-29 10:45:50.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d509ec50bd04cbfca14eafc2e12455a19d3b4959","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-05-29 10:56:52.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/fa7cce6e3a23469aaf2244e7314c73fe : RETRY_LIMIT in 2m 53s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a62a89d14deb4f0f9b698ae8583d31b0 : RETRY_LIMIT in 3m 01s","accounts_in_message":[],"_revision_number":8},{"id":"cd7979042df2c56d1a263b0b9603defbb73ebd54","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2024-05-29 11:01:46.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d5fd8f13a9f9f7d1acee0c08df3fecf9b3ef6138","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-05-29 12:27:00.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3878552436364a708ca1ffb263b6d815 : SUCCESS in 1h 18m 29s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a5acae167e504593815cebe25abaf4bd : FAILURE in 1h 23m 07s","accounts_in_message":[],"_revision_number":8},{"id":"8b32aeb7317d569e73ae6f13100133d7d8cad427","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-05-29 14:21:55.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":9},{"id":"e51969e96db5392ddf164da6cc1346795160057e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 14:35:48.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/018b625e250c4ef696a4bc311309f04d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0d8ef2d9226d4a9b8bd5972134c9925f : RETRY_LIMIT in 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/66a46de7b402420aa6aeae4ea9a732db : SUCCESS in 12m 01s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"37d471fca9ec6a421098364291ba067c6a2e706e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-29 14:44:23.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/3ad0bf9f9ec24d93b2c15f6492b4133d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/66dfb88dcfee494eb5b815789dc81643 : POST_FAILURE in 19m 53s","accounts_in_message":[],"_revision_number":9},{"id":"7a0b8d3c01817a223fc6706cfd0b2503162f912f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-05-29 15:58:45.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/870550 : SUCCESS in 1h 01m 29s","accounts_in_message":[],"_revision_number":9},{"id":"53b80bd06c873d7fef5eee7ae6e312ef0aea385c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 16:53:50.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f1c8d2cfd644ffd9c0b02a9ee1396dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3a76b3df2da40c9b7c5616926fa2823 : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f7c0cec44834ebdb224637efd8c9a12 : SUCCESS in 10m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f996614d1f9149dabccbb1465816aa8d : SUCCESS in 12m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/115bc1a4bd524fd8a871bf5ee4427b71 : SUCCESS in 16m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e4cf7cd936748d980de7ca945332e4a : SUCCESS in 12m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/6bf57c47423846a2a800306e092309cb : SUCCESS in 1h 05m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a54a62bea124dc18d8a24c9250c6449 : SUCCESS in 1h 36m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/058da33844ac4f8bb619b5517ab62256 : SUCCESS in 57m 31s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/919e7ee14988434597e4058323dcd1a7 : SUCCESS in 17m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/faddb366513b4808b23ac3562f843ec6 : SUCCESS in 4m 51s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c6ff9a99a9a24f0995fd4365727b1a84 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ced49dba7e554b278e07f79ca77f3ac4 : SUCCESS in 6m 07s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5ab79ef4a77441fa854a3ea1eb59ebd5 : SUCCESS in 5m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ecf411582616455eae2c27e4a4060ad7 : FAILURE in 45m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f5213a247ed147109b3769d80b7f0b2c : FAILURE in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/480bf5751b254a768948093c32b44e3c : SUCCESS in 2h 20m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0b225227a26a4894a4c28e45fe4b6599 : FAILURE in 2h 26m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b12a5290648641b5869f2a97ac6b2c80 : SUCCESS in 1h 39m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8d88a66162b143b9be88393308187db1 : SUCCESS in 1h 51m 18s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/1a5fd93b172f4899a9a1c5f58dc103fa : SUCCESS in 32m 10s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/b0dc1b8b6a5a4a869cdbed96f0d17755 : SUCCESS in 57m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1e0502eb74f744eebc90f550f0f7413a : SUCCESS in 1h 40m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/66bcb42ea89f4255b5309b39cfa43993 : SUCCESS in 35m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/e430fb503ad24196bec08f874cc23f29 : SUCCESS in 1h 21m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/dc23ae09e02d4eaeac5674d0767aeaf1 : SUCCESS in 1h 22m 27s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f2a1e039b5b241f3801da79f4b17fee7 : SUCCESS in 1h 56m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d3a58154a7945faaadb885f2b45dc55 : SUCCESS in 1h 05m 05s","accounts_in_message":[],"_revision_number":9},{"id":"27b5c2daf237fafa8bdedbcf3061cf22bb636c93","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-05-29 17:00:03.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/29_05_2024_16_54_00\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":9},{"id":"cbf73940ec13ab2804a49c2ea4d5da695564a1a2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-05-29 17:00:38.000000000","message":"Patch Set 9:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/535/.","accounts_in_message":[],"_revision_number":9},{"id":"0f2858e96c0e1e63ec65000d8d62e79062c15be3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-29 17:09:07.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/9/upstream-check/cinder-tempest-plugin-ontap-fcp/30cf3a4/ : RETRY_LIMIT in 3m 44s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/9/upstream-check/cinder-tempest-plugin-ontap-iscsi/802be43/ : RETRY_LIMIT in 3m 26s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/9/upstream-check/cinder-tempest-plugin-ontap-nfs/6b958d6/ : RETRY_LIMIT in 3m 44s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/50/870550/9/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/ca77e82/ : RETRY_LIMIT in 3m 25s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"1a6437045335501906e63287c8050046d30ad9b1","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-05-29 17:25:14.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"97e35dbb60fc6565b522e464b47d6d7106d1241d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 17:25:50.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"3bea7cbb49ceff18ec2e98023898934e0c9ca926","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-05-29 17:35:49.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3fab5f1522c34d8c9336b9ea02357326 : SUCCESS in 1h 25m 06s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/d03c57852b9145a5bf4deb6876fcf166 : SUCCESS in 1h 19m 58s","accounts_in_message":[],"_revision_number":9},{"id":"3478e47dce94b0656c8c1ece0b7310cb1d0c6ce2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 19:28:00.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/90ff1d3f4a4f4214998fea5a1689747c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef963ff5ea8349f4ac45c72f00587008 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c72aac38e57a473eaa14ab74373fa9e5 : SUCCESS in 10m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b3a1078f1ed14290a0069e5785136e51 : SUCCESS in 10m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85100373e56c465497365ec314fdea3d : SUCCESS in 12m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/9e42f1ba77444c6ea9129b66f7936225 : SUCCESS in 1h 00m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4c75503f3c284a03b43407bdead97993 : SUCCESS in 1h 23m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/920ab51f8b644da68ed0866255976e74 : FAILURE in 1h 07m 09s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/58ea434fd6f94ff9bff198010c5057b2 : SUCCESS in 1h 08m 31s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/38d3d88f70bf473aa1e9e02d10d13140 : SUCCESS in 2h 00m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08599d1e96f349039231c008ec71bb7a : SUCCESS in 48m 49s","accounts_in_message":[],"_revision_number":9},{"id":"d69a8e6fe616b3985caaffac4411eba228be7925","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-29 20:37:29.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/50/870550/9/check/DellEMC_PowerStore_iSCSI/3e46beb/DellEMC_PowerStore_iSCSI/1397 : SUCCESS in 1h 24m 33s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/50/870550/9/check/DellEMC_PowerStore_FC/3048d9b/DellEMC_PowerStore_FC/1580 : SUCCESS in 1h 11m 18s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/50/870550/9/check/DellEMC_PowerStore_NFS/ce32c7d/DellEMC_PowerStore_NFS/1091 : SUCCESS in 1h 12m 38s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/50/870550/9/check/DellEMC_PowerStore_NVMeTCP/d8896fb/DellEMC_PowerStore_NVMeTCP/945 : SUCCESS in 1h 53m 44s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"8f9b3f592e4c6977153eef0158a13bede7ce57aa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-29 22:40:20.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/50/870550/9/check/DellEMC_Unity_FC/908964d/DellEMC_Unity_FC/2005 : FAILURE in 1h 29m 19s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/50/870550/9/check/DellEMC_Unity_iSCSI/0b5f5f3/DellEMC_Unity_iSCSI/2030 : FAILURE in 1h 44m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"a8f9513cf0a81505b8bc48dc4857649579c34a2e","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-05-30 03:39:54.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b809ed27125222a86868a2741c4e7bb9ec7302a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 03:52:20.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da8e89f7fbb6401899363aeeceb3987e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a49d904f33a04bd7aaeb0241f93679f0 : RETRY_LIMIT in 1m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/15a32bfa053946d6ab0e987e2d3a309e : FAILURE in 11m 25s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e0bbd7b15b5cec2bbb21fbbae7126f6cfc97c5cf","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-30 03:52:23.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/106e15167c9f48a9a999564cc02916b8\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/643ca96cc1914953a037681f8d3a42e4 : POST_FAILURE in 9m 26s","accounts_in_message":[],"_revision_number":9},{"id":"6bc4ca74af16917259dca9bf4d5be0dc12a62d84","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-05-30 05:04:27.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2ff4bcd9c44a49cb9b52235a5e408344 : SUCCESS in 1h 22m 26s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a75a8d4747c84e969781064af7c39233 : SUCCESS in 1h 17m 59s","accounts_in_message":[],"_revision_number":9},{"id":"6f27d99fe7f9849102ff91e743443f5152ee794d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 06:22:50.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8057ed2884454f2b919cce412efbf276\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8264c6c5fc63463bbbcc09fe0e8371c8 : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6da2f0f6d858483e92ecae35d5af9ac8 : SUCCESS in 10m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3513418a8f2c40008541e336d4aa17ee : SUCCESS in 7m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/16e9905476dc4647b09a38323778668c : SUCCESS in 16m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64f4affb983245ea91353d1bd4019b94 : SUCCESS in 9m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/6d4c1b435ec94cbaa003e80d8928c757 : SUCCESS in 39m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fe99441c792c4348a75478946b9efd2f : SUCCESS in 1h 34m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21b3ce4c266141c39a7233d3025b310a : SUCCESS in 48m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cec01a49d0974437ba354bc0f96cb164 : SUCCESS in 16m 30s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/67b11dc34b814726a3e5e8240ad93d8a : SUCCESS in 4m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e248fd07032d4f008972a0b26e9fe471 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fc5eeecd835b4284b3999c8c39e61445 : SUCCESS in 5m 49s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9d3d345e780d4375a2522a44b13db635 : SUCCESS in 7m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d79d0d263f474b339bc5ac78d4ab6e2e : FAILURE in 38m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/83a38477e79c4665a1c60585a342ed0b : FAILURE in 4m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a23e9085e294492d890bfb636a2764ba : SUCCESS in 1h 59m 33s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a67cbe57a9d74562963c6152bc8f4aa4 : FAILURE in 2h 36m 15s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/623218d636a343d19977362c181e38e3 : SUCCESS in 1h 52m 42s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/043afdb807ce4fb083cc0f8838778df1 : SUCCESS in 1h 51m 14s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/5e9c0402da6e40ffab1022658690ceb3 : SUCCESS in 23m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9a804faf40ba4572a333cb613f9e970c : SUCCESS in 1h 16m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/41734514602b455ca2930014a2ab8554 : FAILURE in 1h 23m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d2926298979849cca15ac7f9d082fcbc : SUCCESS in 46m 25s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f4b3b729dd8043d89cf64ceeafc68dc2 : SUCCESS in 1h 23m 01s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/10bafba77f0a4177adf8e6b888394244 : SUCCESS in 1h 57m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c4913680a4c145f38411f6c3c47d0221 : FAILURE in 1h 43m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3147e9d0a2564f9c95e0acc96ff1783c : SUCCESS in 50m 07s","accounts_in_message":[],"_revision_number":9},{"id":"ce3cf14c2b09748567fda04a28419666559272c2","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-05-30 06:35:06.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e7ecb48321af84c79a46315d777d9ac1afe3fcdd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 06:48:10.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb52602c25814afb802235b71e973f0e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6cf4e9cfeef34cb18099525a44d08836 : RETRY_LIMIT in 1m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9d26277850a34904b1ffacf87a7a326b : SUCCESS in 12m 15s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f97f248989ec711cd764d90339837ed1c5c5aa8f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-05-30 06:59:43.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/a30ad8248c7a4c5cbc0c3cb5e9c21373\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/2d5ae06399f04c38ae46add8a6641a45 : POST_FAILURE in 21m 11s","accounts_in_message":[],"_revision_number":9},{"id":"d724f133308b9329ca1273189450fc44c7118afd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 09:13:23.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4d2b8245c9342d881596f8bc4dc17ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90160f728e6e4b5484123cd60023d724 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/422e7ab62e5b41158ea39b8350bea31d : SUCCESS in 11m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb0d858b1a0c4bf29a8e7b787d9b2363 : SUCCESS in 8m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/57bb86ebf68d4e72a9844f3ca87c8a42 : SUCCESS in 12m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e4794c5e8354f248867b2ea0cdbcae2 : SUCCESS in 10m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/b2bf9cfffbb541eb8989941d53bd01c2 : SUCCESS in 49m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7b2bca1ecf074bc4b59cee09964b1409 : SUCCESS in 1h 09m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df7656f6db3c4a4db117827f4be6c13d : SUCCESS in 1h 07m 02s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/39b82bb384af4e188c985e9d54818c6d : SUCCESS in 13m 04s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/068cb6fac0a841959f5aa65390c0f077 : SUCCESS in 5m 04s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6e07509194644ba58744c27bd032bde9 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/715f6ae17cbb476f80916314fb13f41a : SUCCESS in 6m 25s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b9b41ef333754175a3aaed23451239a0 : SUCCESS in 5m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1e1b2613eefd4541b87856dda2af5892 : FAILURE in 47m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cf7abec1c58a4c44852fb5f53f971291 : FAILURE in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/613f8f961a4e4e36a7345894a7a52741 : SUCCESS in 1h 59m 27s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/e60d4db4f7a14dcc8276fbce2f98128f : SUCCESS in 2h 31m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ebce8ed8d4714da29ab306543682c111 : SUCCESS in 1h 33m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/3fdfce949b9f4416a2421bf0601b186f : SUCCESS in 1h 58m 25s (non-voting)\n- cinder-tempest-plugin-protection-functional https://zuul.opendev.org/t/openstack/build/a4c12b1b79594ee1b23560e183f815d8 : SUCCESS in 23m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/8373673f03c140e3a1f2028be4f86648 : SUCCESS in 1h 09m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0599a05f2cbe491db5c657127a4d7671 : SUCCESS in 1h 31m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/57140d079228443082f958a7d68df82f : SUCCESS in 31m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/72857c3d71c14ce483fcafa78895edd1 : SUCCESS in 1h 08m 46s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/0dd682ca12c44dcbb99f5a8960c18083 : SUCCESS in 1h 36m 36s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cf06cff8d49541289a9552af7343b01d : SUCCESS in 1h 53m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/011e6e3f81854a558ff60088b0fa3d07 : SUCCESS in 1h 03m 26s","accounts_in_message":[],"_revision_number":9},{"id":"bb7d416396901d583a0dae2a26189c918c424c15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 09:13:58.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"9a71fe5cdf7df6352c9ef480abd99d74b564d9c4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-05-30 09:19:41.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_05_2024_09_13_32\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":9},{"id":"a551e8f9fa1046db4286ef920ed756ec345d484e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-30 09:26:40.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp finger://managesf.services.openstack.netapp.com:7979/f9519f8c3b0e435f8e8380c4fbcda215 : RETRY_LIMIT in 1m 42s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi finger://managesf.services.openstack.netapp.com:7979/4fc79d12c9ab4976a173ef52fed17954 : RETRY_LIMIT in 1m 58s\n- cinder-tempest-plugin-ontap-nfs finger://managesf.services.openstack.netapp.com:7979/f6572f399c2c4d408b4b1f6b40dad999 : RETRY_LIMIT in 2m 09s\n- cinder-tempest-plugin-ontap-nvme-tcp finger://managesf.services.openstack.netapp.com:7979/554900c87b4f4bb2a2e9feb3dd9f9024 : RETRY_LIMIT in 1m 43s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"7bd5d2f0068a950942e9b97fef7c400195a09e1c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-05-30 09:38:38.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/26c22381a8e244a78bfeaaec84633fce : SUCCESS in 1h 14m 18s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/17e4fcb019584c5cac632df2de53aba1 : SUCCESS in 1h 16m 08s","accounts_in_message":[],"_revision_number":9},{"id":"0cec3e273b4df1e9de3592ea47fe9b5bd7f9fca5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 11:42:38.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c45f7dc2e5c94f5cb0a8f74549063601\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11539ac2c5ce4e67b102b93f05ecbdad : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c79ed9f90fe643779dc11bd5bd8af699 : SUCCESS in 12m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3149e3f3deb949069fc057ed5a051f47 : SUCCESS in 12m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6975d8dce1c144c5adcb08da7575e4bd : SUCCESS in 12m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/f2ac9d212d2141668ea3f5d8c83d9327 : SUCCESS in 1h 08m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f4ad0a5e50a64b41899ea6aec6fddb5c : SUCCESS in 1h 34m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da0cb6deb2bc4eecbe9ff4c65b5cb57d : SUCCESS in 1h 07m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c23b39b7cf3c4e92936ff9fa3219961b : SUCCESS in 1h 11m 08s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/362e21a250a547d8b03c56118eb02d9d : SUCCESS in 2h 13m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e771702df5f43e99692af90f6c0d917 : SUCCESS in 52m 34s","accounts_in_message":[],"_revision_number":9},{"id":"c5d7f2c329084e1d9c02bed3bddbc3c5acb33a04","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 11:42:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"1505b4b654d8d885d195b2d714f1e37ddd8740f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 11:43:53.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1822f3dc58be4cd2bf7a58ac288ee3bc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4de9c3fa4f974e8e9b0cf6fe318d9242 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":9},{"id":"7f245cd82ce43b1a34b41470d33a5d44103fcc75","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-30 12:05:51.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/50/870550/9/check/DellEMC_Unity_FC/3d1f26b/DellEMC_Unity_FC/2008 : SUCCESS in 2h 10m 54s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/50/870550/9/check/DellEMC_Unity_iSCSI/7ff2995/DellEMC_Unity_iSCSI/2033 : FAILURE in 1h 44m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"8b74a752ce31adfa08b8ecfa7f49b727881e1f40","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-30 13:22:51.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/50/870550/9/check/DellEMC_PowerStore_iSCSI/cb3c82e/DellEMC_PowerStore_iSCSI/1399 : SUCCESS in 1h 18m 17s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/50/870550/9/check/DellEMC_PowerStore_FC/af51b97/DellEMC_PowerStore_FC/1582 : SUCCESS in 1h 04m 57s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/50/870550/9/check/DellEMC_PowerStore_NFS/e382777/DellEMC_PowerStore_NFS/1093 : SUCCESS in 1h 14m 33s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/50/870550/9/check/DellEMC_PowerStore_NVMeTCP/01978ad/DellEMC_PowerStore_NVMeTCP/947 : SUCCESS in 2h 14m 09s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"a2c5a20d30fd2905a35b39e5ef583cc0773bb8bc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-30 16:18:53.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_3.6 https://elab-os-logsrv.delllabs.net/50/870550/9/check/DellEMC_PowerFlex_3.6/df06de1/DellEMC_PowerFlex_3.6/1487 : SUCCESS in 1h 42m 57s\n- DellEMC_PowerFlex_4.5 https://elab-os-logsrv.delllabs.net/50/870550/9/check/DellEMC_PowerFlex_4.5/62433a7/DellEMC_PowerFlex_4.5/401 : SUCCESS in 2h 15m 17s\n- DellEMC_PowerFlex_4.x_NVMeTCP https://elab-os-logsrv.delllabs.net/50/870550/9/check/DellEMC_PowerFlex_4.x_NVMeTCP/7540f79/DellEMC_PowerFlex_4.x_NVMeTCP/340 : FAILURE in 2h 56m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"674ed0c9f0bda8f85bbb6981ad2fc83346a69a22","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-05-30 18:00:08.000000000","message":"Patch Set 9:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/50/870550/9/fc : SUCCESS in 0h 42m 24s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/50/870550/9/iscsi : SUCCESS in 0h 57m 18s","accounts_in_message":[],"_revision_number":9},{"id":"d233ba82cba2b3034962d3280b8fdbfb190112b4","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-06-01 04:16:06.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/870550-9/HPMSA-iSCSI/ : SUCCESS in 3:19:27\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/870550-9/HPMSA-FC/ : SUCCESS in 1:58:52","accounts_in_message":[],"_revision_number":9},{"id":"190616fce652e70a1d9ebee2d71aa302ce5139d0","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-06-01 12:26:21.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/870550-9/Seagate-iSCSI/ : SUCCESS in 2:24:26\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/870550-9/Seagate-FC/ : SUCCESS in 2:17:16","accounts_in_message":[],"_revision_number":9},{"id":"2d4ecc91eae55f41e91de5bdbf4e49f5b479dec6","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-06-01 20:50:56.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/870550-9/Lenovo-iSCSI/ : SUCCESS in 2:24:15\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/870550-9/Lenovo-FC/ : SUCCESS in 2:17:09","accounts_in_message":[],"_revision_number":9},{"id":"ff262587bf2bb39f09ef464bc5bb3043f8f353c5","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2024-06-02 02:50:49.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/870550-9/HPMSA-iSCSI/ : SUCCESS in 3:20:17\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/870550-9/HPMSA-FC/ : SUCCESS in 1:58:31","accounts_in_message":[],"_revision_number":9},{"id":"5a41a8d55109f14bf3e51d769ccf90e995d5e5bb","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2024-06-02 09:53:31.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/870550-9/Seagate-iSCSI/ : SUCCESS in 2:24:59\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/870550-9/Seagate-FC/ : SUCCESS in 2:17:45","accounts_in_message":[],"_revision_number":9},{"id":"d2ca155a0beb64b839cb179910fc3eadc1bb5ad5","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2024-06-02 19:35:58.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/870550-9/Lenovo-iSCSI/ : SUCCESS in 2:27:25\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/870550-9/Lenovo-FC/ : SUCCESS in 2:19:20","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"2ad9fe9156802c41806dbe7975f5f1efeee84675","revisions":{"5726b2fe29845bd3402e995988ee9a5666633c09":{"kind":"REWORK","_number":1,"created":"2023-01-16 14:56:42.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/50/870550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/50/870550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/50/870550/1"}}},"commit":{"parents":[{"commit":"eb309c5d62147485916574593d597633b86e1099","subject":"StorPool driver: remove the obsolete backup_volume()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/eb309c5d62147485916574593d597633b86e1099"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-09-26 13:04:36.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-01-16 14:51:21.000000000","tz":120},"subject":"StorPool: drop copy_image_to_volume() and copy_volume_to_image()","message":"StorPool: drop copy_image_to_volume() and copy_volume_to_image()\n\nThese methods seem to be leftovers from a bygone era when the parent\ndriver could not or would not attach volumes safely.\n\nChange-Id: I6e1a9026e677aee6c7ccad908fe6f92dc253762a\nCloses-Bug: #2002996\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5726b2fe29845bd3402e995988ee9a5666633c09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/5726b2fe29845bd3402e995988ee9a5666633c09"}]},"branch":"refs/heads/master"},"6cba1c8f298bcb1f963b78ddad25f3092dc59cfe":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-01-26 12:35:12.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/50/870550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/50/870550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/50/870550/2"}}},"commit":{"parents":[{"commit":"d5e76e01cbd84dfe339ff50a28b846200965e7c7","subject":"StorPool driver: remove the obsolete backup_volume()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d5e76e01cbd84dfe339ff50a28b846200965e7c7"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-09-26 13:04:36.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-01-26 12:33:35.000000000","tz":120},"subject":"StorPool: drop copy_image_to_volume() and copy_volume_to_image()","message":"StorPool: drop copy_image_to_volume() and copy_volume_to_image()\n\nThese methods seem to be leftovers from a bygone era when the parent\ndriver could not or would not attach volumes safely.\n\nChange-Id: I6e1a9026e677aee6c7ccad908fe6f92dc253762a\nCloses-Bug: #2002996\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6cba1c8f298bcb1f963b78ddad25f3092dc59cfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6cba1c8f298bcb1f963b78ddad25f3092dc59cfe"}]},"branch":"refs/heads/master"},"f52a613914475a19ae79b2b5d622917d94e81b47":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-02-08 13:33:35.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/50/870550/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/50/870550/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/50/870550/3"}}},"commit":{"parents":[{"commit":"1dfd77e3288eaeecd4ce2d2e6bb64871a1f59874","subject":"StorPool driver: remove the obsolete backup_volume()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1dfd77e3288eaeecd4ce2d2e6bb64871a1f59874"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-09-26 13:04:36.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-02-08 13:33:11.000000000","tz":120},"subject":"StorPool: drop copy_image_to_volume() and copy_volume_to_image()","message":"StorPool: drop copy_image_to_volume() and copy_volume_to_image()\n\nThese methods seem to be leftovers from a bygone era when the parent\ndriver could not or would not attach volumes safely.\n\nChange-Id: I6e1a9026e677aee6c7ccad908fe6f92dc253762a\nCloses-Bug: #2002996\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f52a613914475a19ae79b2b5d622917d94e81b47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f52a613914475a19ae79b2b5d622917d94e81b47"}]},"branch":"refs/heads/master"},"a857a90934dd61a5e3d8bbb0644fbc4cf1c72311":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-02-13 09:20:32.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/50/870550/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/50/870550/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/50/870550/4"}}},"commit":{"parents":[{"commit":"f1b90e9a579c6192749ee67950bce68bd7fc3291","subject":"StorPool driver: remove the obsolete backup_volume()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/f1b90e9a579c6192749ee67950bce68bd7fc3291"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-09-26 13:04:36.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-02-13 09:18:14.000000000","tz":120},"subject":"StorPool: drop copy_image_to_volume() and copy_volume_to_image()","message":"StorPool: drop copy_image_to_volume() and copy_volume_to_image()\n\nThese methods seem to be leftovers from a bygone era when the parent\ndriver could not or would not attach volumes safely.\n\nChange-Id: I6e1a9026e677aee6c7ccad908fe6f92dc253762a\nCloses-Bug: #2002996\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a857a90934dd61a5e3d8bbb0644fbc4cf1c72311"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a857a90934dd61a5e3d8bbb0644fbc4cf1c72311"}]},"branch":"refs/heads/master"},"181c3d9d57517ed9e6044a3ccafdf8d6d5cc5247":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-03-01 12:49:08.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/50/870550/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/50/870550/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/50/870550/5"}}},"commit":{"parents":[{"commit":"72a612e7e6680a1ddb4c36696919e1799ece295b","subject":"StorPool driver: remove the obsolete backup_volume()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/72a612e7e6680a1ddb4c36696919e1799ece295b"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-09-26 13:04:36.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-03-01 12:49:01.000000000","tz":120},"subject":"StorPool: drop copy_image_to_volume() and copy_volume_to_image()","message":"StorPool: drop copy_image_to_volume() and copy_volume_to_image()\n\nThese methods seem to be leftovers from a bygone era when the parent\ndriver could not or would not attach volumes safely.\n\nChange-Id: I6e1a9026e677aee6c7ccad908fe6f92dc253762a\nCloses-Bug: #2002996\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/181c3d9d57517ed9e6044a3ccafdf8d6d5cc5247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/181c3d9d57517ed9e6044a3ccafdf8d6d5cc5247"}]},"branch":"refs/heads/master"},"9933e658d478513db6d11f1ce0340a3cb337bdf1":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-05-08 19:32:02.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/50/870550/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/50/870550/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/50/870550/6"}}},"commit":{"parents":[{"commit":"98a619579b8df5336ea6a85a5871c2d4a5db808e","subject":"StorPool driver: remove the obsolete backup_volume()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/98a619579b8df5336ea6a85a5871c2d4a5db808e"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-09-26 13:04:36.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-05-08 19:21:35.000000000","tz":180},"subject":"StorPool: drop copy_image_to_volume() and copy_volume_to_image()","message":"StorPool: drop copy_image_to_volume() and copy_volume_to_image()\n\nThese methods seem to be leftovers from a bygone era when the parent\ndriver could not or would not attach volumes safely.\n\nChange-Id: I6e1a9026e677aee6c7ccad908fe6f92dc253762a\nCloses-Bug: #2002996\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9933e658d478513db6d11f1ce0340a3cb337bdf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9933e658d478513db6d11f1ce0340a3cb337bdf1"}]},"branch":"refs/heads/master"},"682dc2d25fc7f04de77c8ea60e76189d5b714566":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-12-06 06:26:30.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/50/870550/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/50/870550/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/50/870550/7"}}},"commit":{"parents":[{"commit":"a1b26cc52eb2855226052e19301e6e6827dcb699","subject":"StorPool driver: remove the obsolete backup_volume()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a1b26cc52eb2855226052e19301e6e6827dcb699"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-09-26 13:04:36.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2023-12-06 06:19:21.000000000","tz":120},"subject":"StorPool: drop copy_image_to_volume() and copy_volume_to_image()","message":"StorPool: drop copy_image_to_volume() and copy_volume_to_image()\n\nThese methods seem to be leftovers from a bygone era when the parent\ndriver could not or would not attach volumes safely.\n\nChange-Id: I6e1a9026e677aee6c7ccad908fe6f92dc253762a\nCloses-Bug: #2002996\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/682dc2d25fc7f04de77c8ea60e76189d5b714566"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/682dc2d25fc7f04de77c8ea60e76189d5b714566"}]},"branch":"refs/heads/master"},"7daed00eb1208ed62a9bd7f8432ed6d48b2efb01":{"kind":"REWORK","_number":8,"created":"2024-01-18 12:56:47.000000000","uploader":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"ref":"refs/changes/50/870550/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/50/870550/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/50/870550/8"}}},"commit":{"parents":[{"commit":"afb9cdd46314f87969f6f337794406948106fed2","subject":"StorPool driver: remove the obsolete backup_volume()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/afb9cdd46314f87969f6f337794406948106fed2"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-09-26 13:04:36.000000000","tz":180},"committer":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2024-01-18 09:26:08.000000000","tz":120},"subject":"StorPool: drop copy_image_to_volume() and copy_volume_to_image()","message":"StorPool: drop copy_image_to_volume() and copy_volume_to_image()\n\nThese methods seem to be leftovers from a bygone era when the parent\ndriver could not or would not attach volumes safely.\n\nChange-Id: I6e1a9026e677aee6c7ccad908fe6f92dc253762a\nCloses-Bug: #2002996\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7daed00eb1208ed62a9bd7f8432ed6d48b2efb01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7daed00eb1208ed62a9bd7f8432ed6d48b2efb01"}]},"branch":"refs/heads/master"},"2ad9fe9156802c41806dbe7975f5f1efeee84675":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-05-29 14:21:55.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/50/870550/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/50/870550/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/50/870550/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/50/870550/9"}}},"commit":{"parents":[{"commit":"b6236a56a917d8f801bf99205a3552a669bcf143","subject":"StorPool driver: remove the obsolete backup_volume()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b6236a56a917d8f801bf99205a3552a669bcf143"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2022-09-26 13:04:36.000000000","tz":180},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-05-29 14:06:00.000000000","tz":330},"subject":"StorPool: drop copy_image_to_volume() and copy_volume_to_image()","message":"StorPool: drop copy_image_to_volume() and copy_volume_to_image()\n\nThese methods seem to be leftovers from a bygone era when the parent\ndriver could not or would not attach volumes safely.\n\nChange-Id: I6e1a9026e677aee6c7ccad908fe6f92dc253762a\nCloses-Bug: #2002996\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ad9fe9156802c41806dbe7975f5f1efeee84675"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2ad9fe9156802c41806dbe7975f5f1efeee84675"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
