)]}'
{"id":"openstack%2Fcinder~871257","triplet_id":"openstack%2Fcinder~master~If50266d529a39261ec184ee7758f96a3dc89db37","project":"openstack/cinder","branch":"master","topic":"bp/ibm-svf-volumegroup","attention_set":{},"removed_from_attention_set":{"34201":{"account":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"last_update":"2025-06-05 20:38:13.000000000","reason":"Change was abandoned"},"36538":{"account":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"last_update":"2025-06-05 20:38:13.000000000","reason":"Change was abandoned"},"30555":{"account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"last_update":"2025-06-05 20:38:13.000000000","reason":"Change was abandoned"},"35677":{"account":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"last_update":"2023-04-04 15:20:59.000000000","reason":"\u003cGERRIT_ACCOUNT_35677\u003e replied on the change","reason_account":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"}}},"hashtags":[],"change_id":"If50266d529a39261ec184ee7758f96a3dc89db37","subject":"[SVf] Adding support of clone operation for volumegroup","status":"ABANDONED","created":"2023-01-24 06:21:11.000000000","updated":"2025-06-06 07:36:53.000000000","total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8ca02d96fcb5eae6e7b91ac49cded17e2fb35236","_number":871257,"virtual_id_number":871257,"owner":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-13 09:30:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"disliked":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":-1,"date":"2023-04-10 13:39:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-04 15:20:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32465,"name":"IBM Storage CI","email":"hchintha@in.ibm.com","username":"ibm-svf-user"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-24 06:21:43.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-01-24 06:29:31.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-01-24 06:51:06.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-24 06:51:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-24 06:55:56.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-01-24 07:17:53.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-24 07:37:00.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-24 08:57:36.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-01-24 09:09:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-24 12:13:33.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-01-24 13:03:51.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-01-24 13:05:04.000000000","updated_by":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"reviewer":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"state":"CC"},{"updated":"2023-01-24 14:25:08.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-24 14:42:53.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-26 23:20:25.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-01-26 23:24:59.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2023-02-08 15:53:33.000000000","updated_by":{"_account_id":32465,"name":"IBM Storage CI","email":"hchintha@in.ibm.com","username":"ibm-svf-user"},"reviewer":{"_account_id":32465,"name":"IBM Storage CI","email":"hchintha@in.ibm.com","username":"ibm-svf-user"},"state":"CC"},{"updated":"2023-02-09 21:37:44.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-02-13 19:47:02.000000000","updated_by":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"reviewer":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"state":"REVIEWER"},{"updated":"2023-02-15 04:43:38.000000000","updated_by":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"reviewer":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"state":"REVIEWER"},{"updated":"2023-02-16 03:00:59.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-09 12:38:39.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-03-09 14:09:31.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-09 14:33:34.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-03-13 11:09:56.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-13 11:17:32.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-03-13 11:23:19.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2023-03-13 13:16:19.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-04-10 13:39:40.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2024-02-13 06:31:30.000000000","updated_by":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"reviewer":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"state":"REVIEWER"},{"updated":"2024-02-13 13:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"}],"messages":[{"id":"186ffd113a75f50f52f0ab2d9bc1677a6e1acf9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2023-01-24 06:21:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a64d54d249b48ba40b737e88998dfdbef6d3e3b0","tag":"autogenerated:zuul:check","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-24 06:21:43.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"6b11ce0c63bfd27f113714e2017f003c36afaf2e","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-01-24 06:29:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/1500849a1c2f4d12a09ab2e811e94a62 : POST_FAILURE in 7m 38s","accounts_in_message":[],"_revision_number":1},{"id":"f16195595a8830ca8675e93074bcf5b925f7a972","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-24 06:51:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/871257 : FAILURE in 23m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e6844b47010eceaf4452a1fd5b6629402b5b0728","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 06:51:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9d98d9540b84c48b198bfd7b9bd607e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6603b2856f64cd1b07e92b3bd997932 : SUCCESS in 22m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/da048380818a41d38e2054c04527fd6d : SUCCESS in 27m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"076c5b3d6ceefacae6ae0860191a5d2626f7bd87","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-24 06:55:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/57/871257/1/check/iscsi-driver-alletra-6k/ae8c911/ae8c911.html : FAILURE in 25m 07s","accounts_in_message":[],"_revision_number":1},{"id":"22082db930ebd41e3f5546e105342dd444dc9a2e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-24 07:02:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/871257 : FAILURE in 26m 50s","accounts_in_message":[],"_revision_number":1},{"id":"df1ef42fa7b99f73691bcd37de8388f1784f99af","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-24 07:17:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/57/871257/1/ : FAILURE in 48m 20s","accounts_in_message":[],"_revision_number":1},{"id":"5790f129df483f8b66b915a27058b426a19b43db","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-24 07:37:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f70d7f27f24443c8aa21db84e93d8256 : SUCCESS in 1h 13m 48s","accounts_in_message":[],"_revision_number":1},{"id":"939443a96d249d41dc010e771d84c0854f920438","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-24 08:57:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/57/871257/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/b084b24/ : SUCCESS in 2h 07m 27s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/57/871257/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6b07e49/ : FAILURE in 1h 48m 13s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/57/871257/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/48ec402/ : FAILURE in 2h 34m 58s","accounts_in_message":[],"_revision_number":1},{"id":"76e84d766307639e41fe6c55609c4c97291d39a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 09:09:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/38e7aaaaa8b24de88c6b6dbb7fe32bd5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72395f7e70b94810a372c6320227f3f2 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/792aa8000772410fbe806504568ea866 : SUCCESS in 10m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1f2a2612c0c4fd980ee8136418635f6 : SUCCESS in 9m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/833ab8fe60ec4166a65d8f101bacd9ae : SUCCESS in 11m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/b42c722b32b4429db83fc0f36c6756a0 : SUCCESS in 47m 56s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7e110716b2174d5fbb9236b4ea41f257 : SUCCESS in 48m 38s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de95bf512da44b49a78989e9c3746924 : SUCCESS in 1h 20m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e85546083b849129e852189b61a08e8 : SUCCESS in 51m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ea84ec1d06b4b028b10c1cd02b099ec : SUCCESS in 7m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6e52dbc3592b4a14971b009c0132ed3a : SUCCESS in 13m 03s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4f6edfae6e8d46d9a03ec73ec6d50032 : SUCCESS in 4m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/61c221476eb049118b0a0610ff53e665 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5c19aeff851a457f89054580e29548a5 : SUCCESS in 6m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/815be4452d964e65b83308909b2c3001 : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/82cea8a4105847fa86f9010e69a4caf2 : SUCCESS in 36m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2b6ded800d0b4be6b179fbf8dbdd71f3 : FAILURE in 4m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/af5574da2a57431887272e8e0beb3fe5 : SUCCESS in 1h 12m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6eba6cb0d1db46db9d3d0a58846d6eb0 : FAILURE in 1h 24m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b61033a7c39c4b2cb33a000e409665ed : SUCCESS in 1h 19m 33s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4737c978c016448eb4dac2daac5e5373 : FAILURE in 1h 04m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1c40248c46ac49859c9ef38933398ec5 : SUCCESS in 1h 42m 06s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7a73f7522d9a4fe985dff55ea0645b11 : SUCCESS in 29m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7e7f5d56f1b3473f965bbf91a0ac5ba9 : SUCCESS in 1h 21m 50s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/12e4e15056d64d9281651d42c480ef03 : SUCCESS in 1h 38m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/debbabd6ec4b4b96ac4997eef1b41e6f : SUCCESS in 46m 21s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb78a95bf6ea42d2b58f0a6296e0bd02 : SUCCESS in 57m 16s","accounts_in_message":[],"_revision_number":1},{"id":"b9a04c514590439bad72f180bd3e10df5820ddd8","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2023-01-24 09:53:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19dcf4157b1cb6eea10d1733baef7a1cfc87aaa7","tag":"autogenerated:zuul:check","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-24 09:53:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"bc550b514e5f7cca0b3b35c212e44820ae95b45d","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-01-24 09:59:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/33eb277fe990423d82a2283ace2c77b0 : POST_FAILURE in 6m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a29c6f34b0b22096f3893b01f7d231d8f613dc05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 10:20:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/322b816adaf0490e9a121939e06db225\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/42aeea93db3e4773b5f3ca7074d490c7 : SUCCESS in 20m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8d2ef3d5af2e45029f28171716c09e1a : SUCCESS in 23m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24e53a1f59b3006da81974eb92cc47d40c2b0328","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-24 11:07:43.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/87b6b3c9fb314319a61f15e0b3fcdab6 : SUCCESS in 1h 12m 28s","accounts_in_message":[],"_revision_number":1},{"id":"2b871546fe825948dc8efc93c0a918668067a5f7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-24 11:56:34.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/57/871257/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/90d18af/ : SUCCESS in 1h 52m 36s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/57/871257/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0dcc25b/ : FAILURE in 2h 02m 22s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/57/871257/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/a7fc7e2/ : SUCCESS in 1h 38m 21s","accounts_in_message":[],"_revision_number":1},{"id":"bec4d78f1294e6a812d84ab61a5ba98fb7d743fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 12:05:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfbc62f2e55349529f47721e31b278a6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94cd44f84725492f80c6b3499577e43a : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c45bb14d5b74c0cb01a84c4a831557b : SUCCESS in 9m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81e02bd12f8a42d3a52bfeeb710f9db8 : SUCCESS in 10m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eb731ff8ffc447fbc92f96444026b11 : SUCCESS in 12m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/84e7b796dbd34ea987e1ab6d0d8446e2 : SUCCESS in 58m 00s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b3c71391c31d4fe5b78a7e02e5a49904 : SUCCESS in 50m 48s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e2a6c79653b44ecea7f87a3e68592baf : SUCCESS in 1h 33m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed63d3bea74c4c24b25649845e09aedd : SUCCESS in 1h 08m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48d679c39f694a0e8fc4a255e8d19c1e : SUCCESS in 7m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/fc084b5c75bb42c98379c9d94e620055 : SUCCESS in 14m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8dd6240f4e414cea8656a15ade8cdb49 : SUCCESS in 4m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a26348a1fa124a04babedb80ae25c1ee : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5becfe76d83b4138a73e264830bd6b1f : SUCCESS in 4m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8c71ef9309aa44798ba4926a7103faa2 : SUCCESS in 6m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6f6bdadadf134873af4f155ecfc56b2b : SUCCESS in 40m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/93df4f3442c0400ea3b7df8975c9ddaf : FAILURE in 4m 25s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/270b846a89984b03918ba60514b3c6ee : SUCCESS in 1h 19m 09s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4562cc31330f4242813011bbdc989603 : FAILURE in 1h 13m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cbab1a24582b4075a448d1e4e8872c9c : SUCCESS in 1h 43m 56s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e87c68f24efc4da699f7a21f1d3b5941 : SUCCESS in 57m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1a1e4456acce4d23ba20f2f6c012fb2c : SUCCESS in 1h 21m 33s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/edd309b872cb42e690ec7e1ed6a5e147 : SUCCESS in 32m 09s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f2955a178e3145baa7450193dd79658d : SUCCESS in 1h 17m 38s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/76268f8c0f8246918d182e831b4cbd9a : SUCCESS in 2h 06m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c5d3a39caf54eda93d2783cbcf4f657 : SUCCESS in 50m 01s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/913290367588429e996a0526017d9f1d : SUCCESS in 1h 07m 48s","accounts_in_message":[],"_revision_number":1},{"id":"4f6d6ecb0160a0a3bdd03c700d97957d2150741e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-01-24 12:13:33.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/24_01_2023_12_05_17\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"c1ce575035bd0a6a094efe1d57bd786a4e96c6e4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-24 13:03:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/57/871257/1/upstream-check/cinder-tempest-plugin-ontap-fcp/f797ae9/ : SUCCESS in 57m 57s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/57/871257/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/7182244/ : SUCCESS in 42m 58s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/57/871257/1/upstream-check/cinder-tempest-plugin-ontap-nfs/0ddb807/ : SUCCESS in 40m 59s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/57/871257/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/38801d6/ : FAILURE in 30m 21s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"c6feeb1b8febd64ee40b6ed5bf88ade6b36f883f","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-01-24 13:05:04.000000000","message":"Patch Set 1:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/57/871257/1/fc : FAILURE in h m 0s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/57/871257/1/iscsi : FAILURE in h m 0s","accounts_in_message":[],"_revision_number":1},{"id":"4c5749e9156706d90dfd8f0f151c19c0efaad91c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-24 14:25:08.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/8eb7469b868a40b4ad5478fb66402255 : POST_FAILURE in 1h 58m 13s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/2c137ad6316c412da20959e894f36242 : RETRY_LIMIT in 1m 00s","accounts_in_message":[],"_revision_number":1},{"id":"88d334e8ecd748947f4c9c4a4acb743940c58be3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-24 14:42:53.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-57-871257-1/ns5_nfs : FAILURE in 7h 1m 32s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-57-871257-1/ns5_iscsi : FAILURE in 8h 21m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"2e01775907ed92cca00c140dfaf1e3ef663fae1d","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-01-26 23:20:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/871257-1/PVME-iSCSI/ : FAILURE in 1:51:50\n- PVME-FC https://os-logs.seagate.com/dell-ci/871257-1/PVME-FC/ : SUCCESS in 1:35:28","accounts_in_message":[],"_revision_number":1},{"id":"001f2f2b361f204e76020e4e50cd4553b4233777","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-01-26 23:24:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/871257-1/Seagate-iSCSI/ : SUCCESS in 1:54:25\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/871257-1/Seagate-FC/ : SUCCESS in 1:35:30","accounts_in_message":[],"_revision_number":1},{"id":"3a3f0c69a6dfd55f35c6651036e47c2987e8d717","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2023-02-07 12:15:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0bcdd3eca87c4a71ef60ae0eec75c4cbd3567df","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2023-02-08 14:10:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dac6b63336e715d38e48d3da0b720689975308d8","author":{"_account_id":32465,"name":"IBM Storage CI","email":"hchintha@in.ibm.com","username":"ibm-svf-user"},"date":"2023-02-08 15:53:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- storwize-tempest-dsvm-full-iscsi https://ansiblecinder-logs.s3.amazonaws.com/storwize-tempest-dsvm-full-iscsi/d3cf89a/index.html : SUCCESS in 1h 41m 48s\n- storwize-tempest-dsvm-full-fc https://ansiblecinder-logs.s3.amazonaws.com/storwize-tempest-dsvm-full-fc/e3ae60b/index.html : SUCCESS in 43m 05s","accounts_in_message":[],"_revision_number":1},{"id":"9d36e96771a091d61877eb2f074324a77e49d42d","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-02-09 21:37:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/57/871257/1/20230210060853/NEC-V-FC : SUCCESS in 28m 03s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/57/871257/1/20230210060853/NEC-V-ISCSI : SUCCESS in 28m 48s","accounts_in_message":[],"_revision_number":1},{"id":"bd55e412c4056bb6dfbea4cdc5ffe3a799abbd6a","author":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"date":"2023-02-13 19:47:02.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f071bb8a1040c9e3a1842cc89aa60aedce5be0cb","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2023-02-15 04:43:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"09fd41498d5b06d2601d4fb65bb3c6e70bf5e9c9","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-02-16 03:00:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/57/871257/1/20230216045642/NEC-FC : SUCCESS in 3h 31m 48s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/57/871257/1/20230216045642/NEC-ISCSI : SUCCESS in 7h 04m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7662604fe1effc48acddf7e6a6ad29f217f42f8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2023-03-09 12:38:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0c363ffd0e14f069745da1c16d1e3a7a2c8f2ff3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-09 12:38:39.000000000","message":"Patch Set 2:\n\nBuild Started http://openstack-ci.spb.yadro.com:8080/job/openstack-gerrit-trigger-fc/50/ (1/2)","accounts_in_message":[],"_revision_number":2},{"id":"e930065db1983db88b34e73b6859e9bc1b2d2ca9","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2023-03-09 12:41:00.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"113ef5baa087033dff6a4d3bb6f336f1cf035c0e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-09 13:07:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/871257 : FAILURE in 26m 48s","accounts_in_message":[],"_revision_number":2},{"id":"302327515b29a1be5281d18052c3264763f0ca7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 13:18:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19bbdf9b28ac45a8a91b8653e5dc9029\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce78971e4e6a4b17a59c16a54122efab : FAILURE in 37m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0fc7dc7121ef4f52b88ae207a78aa733 : FAILURE in 31m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fc465704d1e418f95285b888ff05e72623a91a6e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-09 13:48:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d7019a367dca4cfab150cfc8fd1fd1aa : SUCCESS in 1h 08m 05s","accounts_in_message":[],"_revision_number":2},{"id":"985c7c2f2004c6d18481e5c96dad2e3d1f8f9c3a","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-09 13:50:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/f7373190b96f471b9fb6c1926afbca48 : SUCCESS in 1h 11m 16s","accounts_in_message":[],"_revision_number":2},{"id":"e27fe7d7675e4837a1e1db0944a851948173a4c5","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-03-09 14:04:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/57/871257/2/ : FAILURE in 1h 17m 49s","accounts_in_message":[],"_revision_number":2},{"id":"9372c19be0c191ec46abb46dfc393e4e0b10ed13","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-03-09 14:09:31.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/57/871257/2/check/fujitsu-eternus-dx-iscsi/ad585a3 : SUCCESS in 1h 30m 21s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/57/871257/2/check/fujitsu-eternus-dx-fc/d7caeed : FAILURE in 46m 23s","accounts_in_message":[],"_revision_number":2},{"id":"76a4bd3cbfd5947b7624787967eaa0f1db084fb6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-09 14:19:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/57/871257/2/check/iscsi-driver-alletra-6k/6abd4cc/6abd4cc.html : FAILURE in 1h 39m 41s","accounts_in_message":[],"_revision_number":2},{"id":"14bd40869fc923eefe8ea92e376ead32afccf632","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-09 14:33:34.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/57/871257/2/check/DellEMC_Unity_FC/0b72480/DellEMC_Unity_FC/2464 : SUCCESS in 1h 34m 52s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/57/871257/2/check/DellEMC_Unity_iSCSI/91d5e92/DellEMC_Unity_iSCSI/2313 : FAILURE in 1h 54m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"766b2df4bc7899695cf97482c452b0f81eb4184d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-09 14:42:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/57/871257/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/84a1eac/ : SUCCESS in 1h 38m 15s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/57/871257/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0b1529f/ : FAILURE in 2h 01m 16s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/57/871257/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/9ef664b/ : SUCCESS in 1h 38m 41s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/57/871257/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/2ff6baf/ : SUCCESS in 1h 43m 36s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/57/871257/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/35fbbb1/ : SUCCESS in 1h 42m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"468dfceead9a7bae1c0abe3b7bc96f758f94b8e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 15:00:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f141c98ad1084d92b5c0938decc39814\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bd72696f2324495a3382d3c280ff375 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97bac7b762044a4fa76fbb6f62c2e7f6 : SUCCESS in 10m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/404f75f72f33493792d7f8f4f44dc932 : SUCCESS in 12m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08b0866944ee498fb961c1e3fe5d0353 : SUCCESS in 12m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/fc984d70312a4ed9a77603dbc4269320 : FAILURE in 52m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/25e2b6ac10254906a6c5988ffe248185 : SUCCESS in 52m 15s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b23bd711ad894483930d7e276881b0e2 : SUCCESS in 1h 15m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d49250d246ea4f56ab564875c60dbe9d : SUCCESS in 50m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a0b07ef60254a198542f68853a00f5a : SUCCESS in 7m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/37d319af927f494db88e30ed1c3832f7 : SUCCESS in 17m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/6e6099c7e65243f4ba0fca4c539c4138 : SUCCESS in 5m 45s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7a960dba788e4d22b16a50d3cf74e0d4 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2260ca83a5184479909477b118c87b2b : SUCCESS in 6m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f60f6bcc944642f4b36e293cf7c212c5 : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/efab0b32065b4b66b5d92402722bad2c : SUCCESS in 38m 16s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/94b9cacf826d43dc8507451378651ee6 : FAILURE in 6m 22s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/635d2987e1974392807ff4eab41815af : SUCCESS in 1h 13m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c872a4dad72a4a5a98dbcb06fc4db108 : FAILURE in 1h 39m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/243fc971b833437c8d898743f4f06c2c : SUCCESS in 1h 43m 14s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5a242e87707b4c21a47b3342abbf8ab4 : SUCCESS in 59m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2ff8a890c28c43abba8ffc537eb2351b : SUCCESS in 1h 27m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cc7b3a8c4ffe4a9d8b7157630b76f82a : SUCCESS in 48m 02s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/6e965dce20cc4733884402fae48c4e0f : FAILURE in 1h 07m 23s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9f6209704b8848a0bdf0d3561678cea4 : SUCCESS in 2h 15m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c12284ba635483692e1f4d35cd382ae : SUCCESS in 53m 20s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e54c03a23f9445b8c9a3ed32fde1886 : SUCCESS in 1h 30m 53s","accounts_in_message":[],"_revision_number":2},{"id":"c9b245f853f60f4a0b3ed56f79ee0069ade8242d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-09 16:31:10.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/14258449caa6466ca988f1c8286d4aa4 : SUCCESS in 2h 23m 23s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/f5d281262c85490db9d2b71917b5a861 : FAILURE in 1h 46m 55s","accounts_in_message":[],"_revision_number":2},{"id":"42f81e75e29d962ea19f08769402afecf19cfaee","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-09 17:23:09.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-57-871257-2/ns5_nfs : FAILURE in 4h 12m 25s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-57-871257-2/ns5_iscsi : FAILURE in 4h 44m 7s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"ce9e00da390ce2cda442e6bebe1f93ef37c95d39","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-09 21:33:31.000000000","message":"Patch Set 2:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/50","accounts_in_message":[],"_revision_number":2},{"id":"05e0ebea130f762ffa7efc5abcfac517343142ed","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2023-03-10 06:35:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e85fafa08cf9cb8b3a43bbfa35d4e49f5783b46f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-10 06:36:07.000000000","message":"Patch Set 2:\n\nBuild Started http://openstack-ci.spb.yadro.com:8080/job/openstack-gerrit-trigger-fc/71/ (1/2)","accounts_in_message":[],"_revision_number":2},{"id":"4903e5a2530b6ef52e07fcc49289562c1a1ddbec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 06:57:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/560458d598834d4c8bf048fe2a744472\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5fe12454b35e49eeb1de943defed0066 : SUCCESS in 19m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/08dcc0808a97479cba42dc18c4e68ec5 : SUCCESS in 19m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c857cd59e48f326e0be65ddc2880fef54a61546f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-10 07:53:59.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c26785ed6db345daaba43030fd66482f : SUCCESS in 1h 16m 59s","accounts_in_message":[],"_revision_number":2},{"id":"4368a9492cb467f79285199e822fcd7a793859ca","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-03-10 08:13:57.000000000","message":"Patch Set 2:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/57/871257/2/check/fujitsu-eternus-dx-iscsi/f9baf63 : SUCCESS in 1h 32m 28s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/57/871257/2/check/fujitsu-eternus-dx-fc/a51c023 : SUCCESS in 1h 38m 12s","accounts_in_message":[],"_revision_number":2},{"id":"e9b4014effad5d62db032b101a00d26e8d5d79d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 08:56:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc39c332d3b74a7abf77fee8cf8afb23\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/168ca30aebaf4d0ab55d6164d66bbb40 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56ce7ae1d26447eca575820f90d78e3e : SUCCESS in 11m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8513613013640f5969c4ebe3230c5a3 : SUCCESS in 11m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db834dfe2c8344c198235c98d3f3129d : SUCCESS in 13m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/b87e8d11b1c94fe69333820f6ef14511 : FAILURE in 1h 03m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c7aaddb1382949ccb315c9f8ccc90c75 : SUCCESS in 1h 03m 56s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/98dd5143dd6b44b7820aed451f60c254 : SUCCESS in 1h 25m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d9857760a03457889b780fa1439cc1c : SUCCESS in 58m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a048a50ba9a4518aba15afbe3efc3f0 : SUCCESS in 7m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cb4ea074ff2f489da78174228da3d82c : SUCCESS in 16m 06s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/41cadebc80014d7d8c6824a0cc886941 : SUCCESS in 5m 10s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a16f3223a1a642dbb4851e937e9483a7 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eac1e9c0e05d4ae0beb93f9ddf39ef3f : SUCCESS in 7m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/037ea37ad8584ab7950e948c6d10c649 : SUCCESS in 5m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/00508386a086424794c1a2e2bde7dd4d : SUCCESS in 50m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b87ca6d8e8334a248f498ab8858afc44 : FAILURE in 6m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f64d92d0b7d740269d2aa22e80b003b1 : SUCCESS in 1h 28m 58s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bce9a1f48edb436bacd817776103866c : FAILURE in 2h 08m 51s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5ea8fb514d6e41b4abd7cf7d821552f0 : SUCCESS in 1h 54m 59s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/605d2f44a89a486a83c6af0b491b3ef2 : SUCCESS in 1h 46m 28s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/5a09ae7c47b34cd0a79fc8e3f3ce2d6e : SUCCESS in 1h 51m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b9d4067038614a41b430ffbd8f1ad401 : SUCCESS in 41m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/be0ea06c6320402fb054533632dc325b : SUCCESS in 1h 25m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/521f37ea7b95440fb38c9f5fdc32722b : SUCCESS in 2h 20m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a08db807e2c84c309790791faaa386e4 : SUCCESS in 53m 41s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6e3ef9058ca478eb54a861222992eba : SUCCESS in 1h 30m 56s","accounts_in_message":[],"_revision_number":2},{"id":"9aff5632500541864b3bb75aeec68c8339f4f57a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-10 09:16:52.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/57/871257/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/26cbc2b/ : SUCCESS in 1h 36m 17s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/57/871257/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/21a3563/ : FAILURE in 2h 39m 32s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/57/871257/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/ee5ebc9/ : SUCCESS in 1h 37m 43s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/57/871257/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/be4106d/ : SUCCESS in 1h 58m 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/57/871257/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/6d2abb6/ : SUCCESS in 1h 51m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"77bf10c60acdf75aad546c1f63ed818516bd5352","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-10 10:23:47.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/9be27d874e3f4bb1853ebddd189d0d71 : SUCCESS in 1h 39m 45s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/ce0317cecbeb411f804b7c26370f403c : SUCCESS in 1h 49m 23s","accounts_in_message":[],"_revision_number":2},{"id":"1445253568759f66ea8293c41963c65f54186a21","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-10 10:49:27.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/f779a0d4f6ad40ff9b9b5754ea2063d7 : SUCCESS in 1h 20m 31s","accounts_in_message":[],"_revision_number":2},{"id":"8d750fea65fa3239c6498debdfb30a8535b1b42a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-10 12:49:42.000000000","message":"Patch Set 2:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/71","accounts_in_message":[],"_revision_number":2},{"id":"65aab21c0464c59140a1ce5d9dcb56bf8726ad90","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2023-03-13 06:55:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dcf97244d9b5b69eb1fc1c589b58c3cac0a9be21","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-13 06:55:27.000000000","message":"Patch Set 2:\n\nBuild Started http://openstack-ci.spb.yadro.com:8080/job/openstack-gerrit-trigger-fc/85/ (2/2)","accounts_in_message":[],"_revision_number":2},{"id":"50962637e595719bf6a5f6e7217247ff8c218073","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 07:25:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/883ea1a7549a4bcb8cf904f7dad7e368\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc709c7d4ce1445097850c05c3819f8e : FAILURE in 27m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/662db44971a84375810107a68f0fb427 : SUCCESS in 24m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3b5b570da89d87111fd069ef1f8aef795174e5ba","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-13 08:07:05.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3ed2509f039247238f3322ee7a21ac7a : SUCCESS in 1h 10m 02s","accounts_in_message":[],"_revision_number":2},{"id":"f0a7ccd6aed99f3ff2b4108247b1058d39b1e118","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-13 08:07:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/e63afeaf84ba49a6b531d7e4cd0dc90e : SUCCESS in 1h 11m 04s","accounts_in_message":[],"_revision_number":2},{"id":"f3e30914fd2f018e0b2018322fd03822df29552d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-03-13 08:35:39.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/57/871257/2/check/fujitsu-eternus-dx-iscsi/968c6f9 : SUCCESS in 1h 34m 22s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/57/871257/2/check/fujitsu-eternus-dx-fc/d496e88 : FAILURE in 12s","accounts_in_message":[],"_revision_number":2},{"id":"86638b46bbc15c4b42b9eca4baed75d364f2b50e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-13 08:56:07.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/57/871257/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/8a05c0f/ : SUCCESS in 1h 34m 56s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/57/871257/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a04c299/ : FAILURE in 1h 57m 52s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/57/871257/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/58056a3/ : SUCCESS in 1h 38m 02s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/57/871257/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/345ca44/ : SUCCESS in 1h 44m 36s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/57/871257/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/14d24ee/ : SUCCESS in 1h 46m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c5d98670b1b5573b56eeed1fc7bd5defa5c74a08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 09:30:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6af1b22f1444679aac1de66dd933bfd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/117d9e29979a461a9967bcec34672687 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dac5e6de207c40158e74eacb103d8d93 : SUCCESS in 9m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1c146dcc1a2482bb93850b309503758 : SUCCESS in 11m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a21834b5d284a56a974b23606e21b24 : SUCCESS in 12m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/d66920773f93414ab2ccf3f561382a74 : SUCCESS in 51m 32s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/45c30ec6e970469ba4804ab4b717fd4b : SUCCESS in 1h 08m 52s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/98dd964a944b4527af177fb998d68a21 : SUCCESS in 1h 21m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f86d219f81cf443fb3bebaad229cbf23 : SUCCESS in 1h 07m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1365b1eaab1a42e5af9946d74169c3c2 : SUCCESS in 9m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/7ecdbe86d80b4dc2880374ebaf2490eb : SUCCESS in 15m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/09620c5d1a2f4e7aa24b76ed2487b93d : SUCCESS in 4m 46s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5f7e43e6350d47109920ba217820af4d : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/16712368882340d39432478d9da287b3 : SUCCESS in 6m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f21e0d65fc014d00bc32b95cf5db2abf : SUCCESS in 5m 27s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2dd71aaaf0e64dc8b3519f94e016b3c6 : SUCCESS in 51m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9ceca3d3ad84483a9c60f91fe31cd1a2 : FAILURE in 5m 35s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cf30c19ea2454a799c7756cd038b12d8 : SUCCESS in 51m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f04761730ae24c4cb97c4891030d0105 : FAILURE in 1h 49m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fcbffe62ebb24ca4abfeafaad74835bf : SUCCESS in 1h 53m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/df2b8179c4924d51b1bf1af74fe65538 : SUCCESS in 56m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/afda013121d54ce3a7a3c9581b9a62dd : SUCCESS in 1h 32m 02s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2958ed442df24d3a9e63063261d682c7 : SUCCESS in 47m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f8c0c75f15c14e23b07bc3a8ac39ecc8 : SUCCESS in 54m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ba791e4519c04d569bafdfca04a19b45 : SUCCESS in 2h 33m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0ad095935f440a8b080cb6232bf32ac : SUCCESS in 55m 16s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad5851a04dc945068e900b78ceec460e : SUCCESS in 1h 18m 18s","accounts_in_message":[],"_revision_number":2},{"id":"18092e1eaa61af62903177773d2acde1775b999f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-03-13 09:38:50.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/13_03_2023_09_31_06\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"56682a885d8be2dd54ff8e66e6db23736ff7b05d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-13 10:20:45.000000000","message":"Patch Set 2:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/1166","accounts_in_message":[],"_revision_number":2},{"id":"fe125bca9e1a9877537c3af66b1d5f1050ed1c47","author":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"date":"2023-03-13 10:53:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2f48446c3f4f2ea30e3a64f91b58185935ec22e9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-03-13 11:07:21.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/57/871257/2/upstream-check/cinder-tempest-plugin-ontap-fcp/7486198/ : SUCCESS in 59m 55s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/57/871257/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/e8169e1/ : FAILURE in 1h 35m 35s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/57/871257/2/upstream-check/cinder-tempest-plugin-ontap-nfs/cd96c0c/ : SUCCESS in 1h 16m 33s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/57/871257/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/99f3984/ : SUCCESS in 1h 12m 46s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"caabe32ac1ec9ac019bd7cc77e8e06e8c952518b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-03-13 11:09:56.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/57/871257/2/check/DellEMC_PowerFlex/13bceff/DellEMC_PowerFlex/1734 : SUCCESS in 1h 38m 40s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"ab92a792df69965c2211c68416204ddb9d6c2b41","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-13 11:12:11.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/72a33946d2374fd2a46a948b85d8b989 : RETRY_LIMIT in 1m 00s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/e5d0b369b3e84a93afdf127a2028726a : POST_FAILURE in 2h 03m 09s","accounts_in_message":[],"_revision_number":2},{"id":"565b90204a9fe7686da107e2529fe3f91c4ccff8","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-03-13 11:17:32.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/57/871257/2/check/DellEMC_PowerStore_iSCSI/cf48d5c/DellEMC_PowerStore_iSCSI/1128 : SUCCESS in 1h 17m 20s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/57/871257/2/check/DellEMC_PowerStore_FC/e65a374/DellEMC_PowerStore_FC/1250 : SUCCESS in 1h 35m 54s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/57/871257/2/check/DellEMC_PowerStore_NFS/26ff448/DellEMC_PowerStore_NFS/767 : SUCCESS in 1h 15m 37s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/57/871257/2/check/DellEMC_PowerStore_NVMeTCP/0b53ee0/DellEMC_PowerStore_NVMeTCP/631 : SUCCESS in 1h 46m 12s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"78748b8dc3e95aa0f208e8bc287952f32504644a","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2023-03-13 11:23:19.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/57/871257/2/check/DellEMC_SC_FC/dd9b5dd/DellEMC_SC_FC/1201 : SUCCESS in 1h 07m 59s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/57/871257/2/check/DellEMC_SC_iSCSI/a5f0667/DellEMC_SC_iSCSI/1181 : SUCCESS in 1h 51m 39s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"f17937baa72ee4f5ac1bcea99e70cc4b5d310940","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-13 11:24:34.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/57/871257/2/check/DellEMC_Unity_FC/7157232/DellEMC_Unity_FC/2481 : SUCCESS in 1h 53m 18s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/57/871257/2/check/DellEMC_Unity_iSCSI/4aa9f1f/DellEMC_Unity_iSCSI/2330 : FAILURE in 1h 21m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"72041d05663453c032a2961caee171ec327a71b0","author":{"_account_id":32465,"name":"IBM Storage CI","email":"hchintha@in.ibm.com","username":"ibm-svf-user"},"date":"2023-03-13 12:23:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- storwize-tempest-dsvm-full-iscsi https://ansiblecinder-logs.s3.amazonaws.com/storwize-tempest-dsvm-full-iscsi/646a745/index.html : SUCCESS in 1h 28m 40s\n- storwize-tempest-dsvm-full-fc https://ansiblecinder-logs.s3.amazonaws.com/storwize-tempest-dsvm-full-fc/2a0507f/index.html : SUCCESS in 51m 30s","accounts_in_message":[],"_revision_number":2},{"id":"f07b429d0ec70d1ef34e68b7011e770a7f0a04f0","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-03-13 13:05:03.000000000","message":"Patch Set 2:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/57/871257/2/fc : SUCCESS in 2h 12m 06s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/57/871257/2/iscsi : SUCCESS in 2h 18m 28s","accounts_in_message":[],"_revision_number":2},{"id":"401fadb47d4172c2c8aaf5419875c495f125aa46","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-03-13 13:16:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/871257-2/HPMSA-iSCSI/ : SUCCESS in 1:37:13\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/871257-2/HPMSA-FC/ : SUCCESS in 1:17:00","accounts_in_message":[],"_revision_number":2},{"id":"4326cd40cc3804aaa3492ea603140d974ac2f94d","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-03-13 13:59:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/871257-2/PVME-iSCSI/ : SUCCESS in 1:42:13\n- PVME-FC https://os-logs.seagate.com/dell-ci/871257-2/PVME-FC/ : FAILURE in 1:35:41","accounts_in_message":[],"_revision_number":2},{"id":"ade4e73987fb8095cd4a7fd1b6fcb1d4160efd37","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-03-13 23:25:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/871257-2/Seagate-iSCSI/ : SUCCESS in 1:43:16\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/871257-2/Seagate-FC/ : SUCCESS in 1:30:14","accounts_in_message":[],"_revision_number":2},{"id":"b676a63592197dda3d37dc7270b0272f0cf9e5f6","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-03-14 17:27:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/57/871257/2/20230314233316/NEC-V-ISCSI : SUCCESS in 2h 16m 37s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/57/871257/2/20230314233316/NEC-V-FC : SUCCESS in 2h 53m 59s","accounts_in_message":[],"_revision_number":2},{"id":"0999196344095ee480086a999f98d28058a88195","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-03-17 13:49:02.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/57/871257/2/20230317192705/NEC-FC : SUCCESS in 3h 09m 51s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/57/871257/2/20230317192705/NEC-ISCSI : SUCCESS in 3h 21m 53s","accounts_in_message":[],"_revision_number":2},{"id":"2bf57c8a205fbd9621095989323f063e0d3d6d8c","author":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"date":"2023-04-04 15:20:59.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"309e8d0fff6f1860cc15f97cb6603653e0379279","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-04-10 13:39:40.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"588156d4fcc19c297b7aaca7b35318eedc877469","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-23 01:42:17.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/57/871257/2/check/iscsi-driver-nimble/e8c0a73/e8c0a73.html : FAILURE in 43m 50s","accounts_in_message":[],"_revision_number":2},{"id":"2553653450ad9505d7556c19628bd535b9bd9506","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-02-13 13:00:07.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/57/871257/2/fc : SUCCESS in 0h 54m 27s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/57/871257/2/iscsi : SUCCESS in 1h 05m 21s","accounts_in_message":[],"_revision_number":2},{"id":"812e25222b57a3d0446a7791de0b980b7ff6f3f8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-06-05 20:38:13.000000000","message":"Abandoned\n\nAbandoning per request on ML.","accounts_in_message":[],"_revision_number":2},{"id":"8ca02d96fcb5eae6e7b91ac49cded17e2fb35236","author":{"_account_id":36538,"name":"Harsh Ailani","display_name":"Harsh Ailani","email":"harsh.ailani@ibm.com","username":"harshailani"},"date":"2025-06-06 07:36:53.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ff0b174e8866d22021aad92d2a7898116139c862","revisions":{"6bd57dd4b8903258692fd043b0f096c123229c0b":{"kind":"REWORK","_number":1,"created":"2023-01-24 06:21:11.000000000","uploader":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"ref":"refs/changes/57/871257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/57/871257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/871257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/871257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/871257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/57/871257/1"}}},"commit":{"parents":[{"commit":"4c9b76b9373a85f8dfae28f240bb130525e777af","subject":"Merge \"Fix Infinidat driver consistency groups feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4c9b76b9373a85f8dfae28f240bb130525e777af"}]}],"author":{"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","date":"2023-01-24 06:18:36.000000000","tz":0},"committer":{"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","date":"2023-01-24 06:18:36.000000000","tz":0},"subject":"[SVf] Adding support of clone operation for volumegroup","message":"[SVf] Adding support of clone operation for volumegroup\n\n[Spectrum Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup and clone of a\nvolumegroup-snapshot operations.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: If50266d529a39261ec184ee7758f96a3dc89db37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6bd57dd4b8903258692fd043b0f096c123229c0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6bd57dd4b8903258692fd043b0f096c123229c0b"}]},"branch":"refs/heads/master"},"ff0b174e8866d22021aad92d2a7898116139c862":{"kind":"REWORK","_number":2,"created":"2023-03-09 12:38:32.000000000","uploader":{"_account_id":34201,"name":"Mounika Sreeram","email":"sreeram.mounika@ibm.com","username":"sreerammounika"},"ref":"refs/changes/57/871257/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/57/871257/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/871257/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/871257/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/57/871257/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/57/871257/2"}}},"commit":{"parents":[{"commit":"3e5ba7a6119a420c04741b60d79f910f2449ab13","subject":"Merge \"Configure a storage node in cinder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3e5ba7a6119a420c04741b60d79f910f2449ab13"}]}],"author":{"name":"sreerammounika","email":"sreeram.mounika@ibm.com","date":"2023-03-09 12:33:39.000000000","tz":0},"committer":{"name":"sreerammounika","email":"sreeram.mounika@ibm.com","date":"2023-03-09 12:37:00.000000000","tz":0},"subject":"[SVf] Adding support of clone operation for volumegroup","message":"[SVf] Adding support of clone operation for volumegroup\n\n[Storage Virtualize family] As part of Flashcopy 2.0 implementation,\nadded support for clone of a volumegroup and clone of a\nvolumegroup-snapshot operations.\n\nImplements: blueprint ibm-svf-volumegroup\nChange-Id: If50266d529a39261ec184ee7758f96a3dc89db37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ff0b174e8866d22021aad92d2a7898116139c862"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ff0b174e8866d22021aad92d2a7898116139c862"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
