)]}'
{"id":"openstack%2Fcinder~871790","triplet_id":"openstack%2Fcinder~stable%2Fwallaby~I7527a7493752c1b3db774de989250ec5b514237b","project":"openstack/cinder","branch":"stable/wallaby","topic":"grenade-drop","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-01-27 16:01:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7527a7493752c1b3db774de989250ec5b514237b","subject":"[stable-em only] move grenade jobs to experimental","status":"MERGED","created":"2023-01-25 21:39:58.000000000","updated":"2023-01-27 18:23:12.000000000","submitted":"2023-01-27 16:01:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871790-grenade-drop","meta_rev_id":"b9daa2802838dd2f66ab4bc8de9a7143ef98c051","_number":871790,"virtual_id_number":871790,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-27 16:01:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":2,"date":"2023-01-27 14:11:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-27 18:23:12.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2023-01-26 18:10:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":1,"date":"2023-01-27 14:11:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-25 21:40:52.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-01-25 21:45:03.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-01-25 21:52:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-25 22:04:52.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-25 22:06:57.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-01-25 23:05:01.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-26 00:03:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-26 04:38:06.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-26 06:04:20.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-26 18:10:45.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-01-27 14:11:43.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2023-01-27 18:23:12.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"526f867e26a971e09db09a2c063a8d587f032cfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-01-25 21:39:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1185dbc48caf09ad0ce6db6e2325b6b40f445f6d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-25 21:40:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"a0f67bb386ee8c6dfb7a05e3e6027731297c5d19","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-25 21:45:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9fe02edad8cf3d12cf4965a4db9fb8e7db740d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 21:52:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/353759f2802c4661885e5a7283f3f7ae\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/38194d6122334462a702d85a3e5d149d : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3d27dfbb3a52471ebf4f2bc50d77b347 : SUCCESS in 10m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6504632eb92551fadf0b589fb91f697ad59f4b17","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-25 22:04:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/871790 : FAILURE in 23m 25s","accounts_in_message":[],"_revision_number":1},{"id":"8dc9725c1a50775ca93cb6ff7243677408466de5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-25 22:06:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/90/871790/1/check/iscsi-driver-alletra-6k/8de1b1c/8de1b1c.html : FAILURE in 25m 57s","accounts_in_message":[],"_revision_number":1},{"id":"935be7298d413bf5c084c1d63dd351a51c4e6aac","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-25 22:07:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/871790 : FAILURE in 26m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8f8d368e4f3e2e2e5bc09d02f404b68f9fa08e49","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-25 23:05:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/90/871790/1/ : FAILURE in 1h 16m 15s","accounts_in_message":[],"_revision_number":1},{"id":"01ee760ac5a96417a6c415a95af14e8658e46333","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 00:03:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d809d639774454fa6e6dfd2753fc689\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26dfcbcb4d3246b69a6f98cb6300e819 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fdd4276104074649a156bc6b7c14364d : SUCCESS in 12m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b21c324518d64d769b6ad9a05931fc3d : SUCCESS in 10m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/537ed86f1fc2486299aef36430a67ce6 : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f52cc3a3455b499288ee741484684eb3 : SUCCESS in 12m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/2061ca1608914e49ac0c1f122985983c : FAILURE in 56m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9b9d13a38ddd4368ba0fe25d7f0722e3 : SUCCESS in 1h 04m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a5d3c76a1d14b06a1a4d392a70b533b : SUCCESS in 57m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f9f999a9674f4f8196abc941a102cab9 : SUCCESS in 11m 45s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/95eb51c535564008b2f02766cd9d08a1 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c9b25fea87a24167abf6d13779df04b9 : SUCCESS in 6m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7f5261512b2e4e31888d76efa286a49f : SUCCESS in 7m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7427a76ffcb54df9af7d83736d573242 : SUCCESS in 32m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3c4e5f9c0aff43dea661b91858a7a001 : SUCCESS in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/394f4d2defb54b0eb940a320bc731f70 : SUCCESS in 1h 17m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/1a33b945e8cd4db9a0f48e101557b613 : FAILURE in 1h 24m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a226b981ce0b494ca22ac4d0d5d05077 : SUCCESS in 2h 01m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0a0a125140444e4bba919d335dd2f479 : SUCCESS in 1h 43m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/7260461ef2174cf79725563bcee88917 : SUCCESS in 1h 11m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fa1b080abdcd4e6fb695892df9b9a673 : SUCCESS in 2h 19m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cea1f6214f14cb68db3699dd631966c : SUCCESS in 1h 06m 30s","accounts_in_message":[],"_revision_number":1},{"id":"00de9e02b58800e9ed6f6071cef6ea2a320db125","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-26 04:38:06.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/8bc3233873c64b5ba21c44640d35df38 : POST_FAILURE in 1h 19m 13s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"9da7f2d4c66288920448d8329e5e39a0ce51dbcf","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-26 06:04:20.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-871790-1/ns5_nfs : SUCCESS in 8h 21m 56s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-871790-1/ns5_iscsi : SUCCESS in 8h 23m 53s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"5188cc32bf651aaae7d3fdb7ea6e17befe6fe418","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-01-26 14:06:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58454ed49e5092c2f5da0fc9339469ccde6f5b0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-01-26 14:12:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2c9d495bc2162832e95814a53f2e16740c0e7ef4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-26 14:14:19.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"4557af68795f445f63afaecf7e322cd654d4b26b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-26 14:17:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3ad32b0e18a84d271d23c2586252dab0fcf4747c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-01-26 14:20:33.000000000","message":"Topic grenade-drop-w removed","accounts_in_message":[],"_revision_number":2},{"id":"5a946827d397edf41e4723d971931f7ff3f24ac4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-01-26 14:20:44.000000000","message":"Topic set to grenade-drop","accounts_in_message":[],"_revision_number":2},{"id":"743ef7bc8f7740f06db48ba2201f01215af18fa2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 14:40:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a83ed79c73c4d72bcbc2e369b97d152\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc56642537da430d80850306fe92ba4c : SUCCESS in 19m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7b057c1f4c96497a9fd38c74a9dfd0ec : SUCCESS in 25m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b826b92e8c74dfd416f4b69e9b2cb0d96bea2242","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-26 15:45:32.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/871790 : FAILURE in 23m 34s","accounts_in_message":[],"_revision_number":2},{"id":"a8710ab03d20db58057a21dd315e0f39d89a4d9a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-26 15:56:05.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/90/871790/2/check/iscsi-driver-alletra-6k/2815911/2815911.html : FAILURE in 26m 08s","accounts_in_message":[],"_revision_number":2},{"id":"8f2b5a48db366da99478cbdbd371dea7ba659026","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-26 16:21:02.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/871790 : FAILURE in 26m 28s","accounts_in_message":[],"_revision_number":2},{"id":"27104378ad25d22a52160c772f6e34bcbceb9ade","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 16:49:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a4cc4be48024996923ef48346f24bdf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9563311f4324133a7e59a5949ead67c : SUCCESS in 8m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/984e5d7b4ee54cfbaca17f58b4c06087 : SUCCESS in 8m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17c382fd91c94b2c9a359fcc7c2d2510 : SUCCESS in 8m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca455012ab9d4f10ab07f95c7188285c : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6826492ff9a1467993afdbfae3414fc7 : SUCCESS in 11m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cb8e92cbeb4d4bbda0acc9ce02b54977 : SUCCESS in 14m 57s (non-voting)\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/85baa34f566c4d4eb42510e39c1cc45c : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2d9f192d1b4e4dadad1bb12e53d9ffbf : SUCCESS in 5m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aced8970245d4385baf8cf861f31c473 : SUCCESS in 4m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2e37e10e0ad345ce81c5f126e0dd83b6 : SUCCESS in 34m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0bbf97cc1fb145a880fd41e6cfaa3af7 : SUCCESS in 4m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/023d34b896a0448b8f6341250c219aff : SUCCESS in 1h 33m 23s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fc91718ce5bc4ca7bac4507abcb2f228 : SUCCESS in 1h 46m 30s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f369599cf0c24f09bcc898ded762f264 : SUCCESS in 1h 42m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1618cf620a1c4619a633c5c474f4ee93 : SUCCESS in 1h 22m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/209ff08353fe433fa405be09685187a9 : SUCCESS in 1h 08m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aaddb423f8484de08f71edb6439cc7db : SUCCESS in 2h 30m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3cdb4cc2eab7454f9215943c3062d74e : SUCCESS in 1h 18m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/046c2bab35e847f6bbeafe339a7b10f0 : SUCCESS in 56m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b963015c30a5490dbfe718e5b966a875 : SUCCESS in 52m 42s","accounts_in_message":[],"_revision_number":2},{"id":"cc16e9b45fd45194377c22d02834a378c93ae6b6","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-01-26 18:10:45.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"95e5454f57e7b376bf3c9fc09edf9e0aa75024c1","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-26 21:54:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/90/871790/2/ : FAILURE in 1h 12m 46s","accounts_in_message":[],"_revision_number":2},{"id":"0a045d0ee5bee4ec1712cc1d587cbc77dd0086ae","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-26 23:31:48.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/9d0004d941d14f99abfcb357eebc7d53 : POST_FAILURE in 1h 09m 19s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/63cba43d698c44878fe06e453c5b2530 : POST_FAILURE in 1h 19m 35s","accounts_in_message":[],"_revision_number":2},{"id":"b294266be5c4d560bb8e72eb3906fb6d6a2633f2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-27 02:04:40.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-871790-2/ns5_nfs : FAILURE in 10h 28m 25s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-871790-2/ns5_iscsi : SUCCESS in 10h 16m 19s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"c2dd68282234e9f6202ad4183a71b1d7f4733d03","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-01-27 14:11:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4d3f511eca03cc8a130959d29d471c5f546d38aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 14:12:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2698f7dbfbf1ca8ee9efd081611f16281150bebe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 16:01:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10ce7e822a364c9498d246fb0e9c1cab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1c930eed927467b89907af0d3020dd5 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/965786e86a6e43c7b96fbe33997ad4cd : SUCCESS in 9m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fad93381246a4ed990d7bc96ae8daf25 : SUCCESS in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f6869b2511f4a8e8c47ff97369ae338 : SUCCESS in 11m 38s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9783a9a12c5341f3be79dbe9f1edcc58 : SUCCESS in 1h 27m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4a19c389d8b3484ca646ca748b6919a8 : SUCCESS in 1h 01m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19857a7a2add47949ba06cfec4ce22c1 : SUCCESS in 1h 12m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a960650d1fa429a92639a3406e5800e : SUCCESS in 44m 37s","accounts_in_message":[],"_revision_number":2},{"id":"6b2df2bf326207c64af68628a8f9115b7641b1bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 16:01:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4c66e7499a6b842fe783abb1d10f11d1c7048518","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 16:04:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cf1261ff0a24191b290f5e2db783f16\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0f51a8af8a048efa1c0ff3c3658dc75 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2},{"id":"b9daa2802838dd2f66ab4bc8de9a7143ef98c051","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-27 18:23:12.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4cdd18b42925e88b9d777373c0c73dff32aa78ff","revisions":{"e966c3bdc4055615f63879653294bf6a46d74e8a":{"kind":"REWORK","_number":1,"created":"2023-01-25 21:39:58.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/90/871790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/871790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/871790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/871790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/871790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/871790/1"}}},"commit":{"parents":[{"commit":"60a5de1f6df78b04d5b9e8438b506d6651eaefd8","subject":"Use the json format output of qemu-img info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/60a5de1f6df78b04d5b9e8438b506d6651eaefd8"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-01-25 21:33:14.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-01-25 21:33:14.000000000","tz":-300},"subject":"[stable-em only] move grenade jobs to experimental","message":"[stable-em only] move grenade jobs to experimental\n\nGiven that \"upgrades from Extended Maintenance are not supported\nwithin the Community\" [0], move the grenade jobs to the experimental\nqueue for the stable branches in EM status.\n\n[0] https://docs.openstack.org/project-team-guide/stable-branches.html#extended-maintenance\n\nChange-Id: I7527a7493752c1b3db774de989250ec5b514237b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e966c3bdc4055615f63879653294bf6a46d74e8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e966c3bdc4055615f63879653294bf6a46d74e8a"}]},"branch":"refs/heads/stable/wallaby"},"4cdd18b42925e88b9d777373c0c73dff32aa78ff":{"kind":"REWORK","_number":2,"created":"2023-01-26 14:12:50.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/90/871790/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/871790/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/871790/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/871790/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/871790/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/871790/2"}}},"commit":{"parents":[{"commit":"60a5de1f6df78b04d5b9e8438b506d6651eaefd8","subject":"Use the json format output of qemu-img info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/60a5de1f6df78b04d5b9e8438b506d6651eaefd8"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-01-25 21:33:14.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-01-26 14:11:03.000000000","tz":-300},"subject":"[stable-em only] move grenade jobs to experimental","message":"[stable-em only] move grenade jobs to experimental\n\nGiven that \"upgrades from Extended Maintenance are not supported\nwithin the Community\" [0], move the grenade jobs to the experimental\nqueue for the stable branches in EM status.  Additionally, remove\nthe integrated-gate-storage template because it includes grenade.\n(The other jobs in that template were already configured in check\nand gate in our .zuul.yaml, so they continue to be run.)\n\n[0] https://docs.openstack.org/project-team-guide/stable-branches.html#extended-maintenance\n\nChange-Id: I7527a7493752c1b3db774de989250ec5b514237b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4cdd18b42925e88b9d777373c0c73dff32aa78ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4cdd18b42925e88b9d777373c0c73dff32aa78ff"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
