)]}'
{"id":"openstack%2Fcinder~871792","triplet_id":"openstack%2Fcinder~stable%2Fussuri~Iee29e0e980faff341418a4f3c21a6ab91fe894dc","project":"openstack/cinder","branch":"stable/ussuri","topic":"grenade-drop","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-02-15 02:40:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iee29e0e980faff341418a4f3c21a6ab91fe894dc","subject":"[stable-em only] move grenade jobs to experimental","status":"MERGED","created":"2023-01-25 21:46:36.000000000","updated":"2023-02-15 02:44:51.000000000","submitted":"2023-02-15 02:40:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871792-grenade-drop","meta_rev_id":"1b93722f6ef8637a1e53388ad5f8587f6ae9c3ce","_number":871792,"virtual_id_number":871792,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-15 02:40:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":2,"date":"2023-02-01 19:45:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-14 19:05:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-14 19:05:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-25 21:47:28.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-01-25 21:49:20.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-01-25 22:32:21.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-25 22:36:25.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-01-26 00:25:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-26 01:26:53.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-26 02:14:39.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-26 17:19:53.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-01 19:45:16.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2023-02-14 19:05:04.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"}],"messages":[{"id":"f766a824aeb93785bd676f0991fbfb5a51e27c4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-01-25 21:46:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0736869949b355ab6cff66c36d768d56763092c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-25 21:47:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"b27d40d2aaf88949f513c6e6ac0c4bad59473f74","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-25 21:49:20.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio pure-devstack-cinder-tempest-iscsi-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-fc-aio pure-devstack-cinder-tempest-fc-aio : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db30df9a67374fac6859ccd0d3b57e238e215b63","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-25 22:32:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/871792 : FAILURE in 23m 28s","accounts_in_message":[],"_revision_number":1},{"id":"0c123c3230ed9a7936908205b201e710b1538caf","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-25 22:36:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/92/871792/1/check/iscsi-driver-alletra-6k/230c90c/230c90c.html : FAILURE in 25m 59s","accounts_in_message":[],"_revision_number":1},{"id":"bc948787269b3caa539120ad653e68d0e18d84b5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-25 22:38:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/871792 : FAILURE in 26m 26s","accounts_in_message":[],"_revision_number":1},{"id":"3a94e8a891cd3e869e243e84609273c9368f8e2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 00:25:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aaa19d43714a4f5cb920ad2136d3faf0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/477e395edb934ec59574b1e21a7a608e : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e58d01d51b4424bb8ab1ccf36f58647 : SUCCESS in 7m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c61d7905ac9f4c479b946cca350a9f50 : SUCCESS in 12m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f253e9d35a4747868d1132de6dac51e2 : SUCCESS in 13m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/6939e584394b41dabde09032e2b38c53 : FAILURE in 9m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0c9fa7ed5bbe46b39b61dfd755384a93 : SUCCESS in 1h 12m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f5da7fd6fa2412fa47536e147da1c36 : SUCCESS in 45m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/de76521cf5ba4201ac8f87db1af7d539 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dc7dcf1bcd3c44e2a9f64cb4416501b8 : SUCCESS in 5m 42s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/73c799c86abf4e718291bf169cff443d : SUCCESS in 6m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/39a1257e3ede4239bee59fd47a4c1a69 : SUCCESS in 40m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/05312ad16bd6415ea050aa1ba125c7a7 : SUCCESS in 6m 31s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fad7a30b12244536b772fbf037fde394 : SUCCESS in 1h 11m 01s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/46ee51831ff2482a82f5e67b670b3832 : SUCCESS in 1h 13m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/433a775a81a044f78632907b2256b304 : SUCCESS in 1h 26m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d05408fec1ee44d1b971ff7491fc77fb : FAILURE in 58m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/501a1297b92544b79e50bcf69a33c87f : SUCCESS in 2h 33m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b567fe482a454715931685434f48c821 : SUCCESS in 55m 51s","accounts_in_message":[],"_revision_number":1},{"id":"1389fa0911cd3dc285e372afb3d0d2c154618032","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-26 01:26:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/92/871792/1/ : FAILURE in 1h 03m 10s","accounts_in_message":[],"_revision_number":1},{"id":"03d011631d11f53287a9e0f74e99ef14b8fe7924","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-26 02:14:39.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-871792-1/ns5_nfs : FAILURE in 3h 49m 2s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-871792-1/ns5_iscsi : FAILURE in 1h 44m 11s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"cc5ae3b9d067b4848e2da3415eae9a981e9e1218","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-26 17:19:53.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/135a726f40ff49319ec1c47493a9168b : POST_FAILURE in 1h 57m 16s\n- devstack-plugin-ceph-dsvm-bionic-py3 finger://zuul-v3-executor.novalocal/99c10f4d18c2446a82b4924cc3a5054e : RETRY_LIMIT in 1m 21s","accounts_in_message":[],"_revision_number":1},{"id":"8897b354aaa4809c237d99bae96f7874c65352be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-01-26 22:15:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"923c1e44c2e07096bf882823ab415cb46271b779","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-26 22:17:42.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"3691c4ce5202f2a4398613e1aff867b6f4717d84","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-26 22:41:22.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/871792 : FAILURE in 23m 30s","accounts_in_message":[],"_revision_number":2},{"id":"c4be0819ffb0dcfd4655c6c483d3aa291c3f4227","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-26 22:43:18.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/92/871792/2/check/iscsi-driver-alletra-6k/7dde7c6/7dde7c6.html : FAILURE in 25m 59s","accounts_in_message":[],"_revision_number":2},{"id":"fa4fbc33c761972b6e484ee6f070d2f352516687","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-26 22:44:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/871792 : FAILURE in 25m 57s","accounts_in_message":[],"_revision_number":2},{"id":"c3ec4ff2cea1159d041aaf93ed20023cc07ea49c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 00:12:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/128316fe71194f3fa904636bcfde7b71\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19a7986223624f5c8d5e3a73f9b41fdd : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5708c94a409c4d798b4c399ecee55cdb : SUCCESS in 9m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f44d874e5cd54a6f9365fd7ccae6a8fa : SUCCESS in 11m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ac0e344f258438281ecc6848e6442f7 : SUCCESS in 13m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/edffce4cef6d4daabee274cbbc51af90 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ca3c181fe1ea41248827a74f86bd1bbc : SUCCESS in 5m 10s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/cf72d912aef643d8bf696fbaa02ce316 : SUCCESS in 6m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4f7906486ee54e02bc8a8ad5f55e41d1 : SUCCESS in 44m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8280696f181c44a9b33dd3d10753065c : SUCCESS in 3m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ee0967c8dd8e498aad092b8068e4bfea : SUCCESS in 1h 04m 23s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8f39d3b52cb84ed6a89c89d6d27d8bfe : SUCCESS in 1h 35m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3bb447868021457da44c0d588d8e5c5d : SUCCESS in 1h 03m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2ea87947a64b4ac680996b368961a578 : FAILURE in 1h 18m 10s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c15eb6b8379f433093916e9802366ac1 : SUCCESS in 1h 49m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/30d1240b654f412a971836efafe09e0e : SUCCESS in 53m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90af6cabd1f94a5ba4de1b3b123cb170 : SUCCESS in 59m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b916d47cd654b30bb5a423fa4c1bc57 : SUCCESS in 54m 00s","accounts_in_message":[],"_revision_number":2},{"id":"1648f4fe490406b809737f2dae5778d567beab54","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-27 00:24:55.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-871792-2/ns5_nfs : FAILURE in 1h 46m 4s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-871792-2/ns5_iscsi : FAILURE in 1h 45m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"4f03405f3397202bc48cf5bf91cdb7bccc42d2f2","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-27 09:26:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/92/871792/2/ : FAILURE in 1h 02m 42s","accounts_in_message":[],"_revision_number":2},{"id":"f10fddef91fed09054f6464de7cf5c2f29429abf","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-01-27 15:12:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cbe7f4841c164523538359096128bf3ce58b1fa9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-01-27 15:14:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b09da663faecc7dc189f31108cb0120c203983f4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-27 15:20:42.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"f0f940716b9c189812e154f98026a881b623ffff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 17:42:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cde8598233754ef3b6b37b43eebfa171\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71acb52559ec4c6995df9ef0ed6aa4e2 : SUCCESS in 8m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5e4e24a0e234edaa7d94784befa95e9 : SUCCESS in 8m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fadca35cbdfc48258c0173bbfccb6a89 : SUCCESS in 12m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15f32cab6ade46bab251ab9f63af98a2 : SUCCESS in 14m 08s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/85696b0697794fac877a4f394d441d18 : SUCCESS in 11m 31s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/113442fdb9c44906951f2a6ab095e6cf : SUCCESS in 5m 36s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/4d9850e4eeec4fc9b45d324cf3978443 : SUCCESS in 6m 08s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/be2920d6482042bea1b6241784b66ad3 : SUCCESS in 43m 02s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/36687721eb18461a9382be9bfe5d0c55 : SUCCESS in 7m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0a3a0ae3755948fc93e0cda42e614b82 : SUCCESS in 1h 08m 19s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fd9262cdd1dd4d3e98e0e6bd3e9b4732 : FAILURE in 1h 15m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a75c35a5e88144b19c2b91955d5e4f90 : SUCCESS in 1h 37m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/35192445669c45ccac1615ef5308270b : FAILURE in 1h 13m 58s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/893b87f3d54a43acbd9c330cec51465f : SUCCESS in 2h 23m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7abe1cd54dd4431985523dbc8ee7c706 : SUCCESS in 1h 14m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/626059b3cbb24e25a719cc9de9162453 : SUCCESS in 1h 08m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a001316f32514b71ae35df40aa7881e4 : SUCCESS in 52m 07s","accounts_in_message":[],"_revision_number":2},{"id":"ab4b05b464278807e987696295eb6af12f5c51e0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-28 04:55:11.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/378b4822c19d467eaa188557e2108cb2 : RETRY_LIMIT in 1m 05s\n- devstack-plugin-ceph-dsvm-bionic-py3 finger://zuul-v3-executor.novalocal/100ef0d6fbd5457cb5ab95a3dbe23866 : RETRY_LIMIT in 1m 00s","accounts_in_message":[],"_revision_number":2},{"id":"fddcce35222ee452671c9fe4fa4dcb503ca452b9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-02-01 13:02:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a208e2ff21c4f3f6a8bc540dec122684812eeb76","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-02-01 13:05:48.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"04adfc432e6511197755c1c210eef05b407a9120","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 15:53:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3541baf0d160410091387080e4e18977\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ca51c5a912748dbbb160853ef26e707 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f3b7996b52944ae806c09b02c14e9a6 : SUCCESS in 9m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ba0c7ea350644e0b8bbf3003ef18f90 : SUCCESS in 10m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27baea60e2a24dfc81d22fdd5c7173a4 : SUCCESS in 13m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ffd38650a5364c24b851d59e3bcc2c56 : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6dde782da12c4204aff65f23ec88385a : SUCCESS in 5m 36s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c132f38bbb8545fb9dbf265f8713589a : SUCCESS in 6m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/be77a8b8f8ef49e0ab86e45f95e47313 : SUCCESS in 44m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/25a9fd8d197e4aeabe6803c62cc9add3 : SUCCESS in 4m 55s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/31ed133fd86047579553d79fb26d464a : SUCCESS in 1h 13m 43s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cdad919a5daa43f08e4cc93685941a83 : SUCCESS in 1h 31m 15s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/eb71b5135e1248ac8345d23934c90a29 : SUCCESS in 1h 28m 07s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0f6baa2527cc44febce976d87ac12963 : FAILURE in 1h 00m 49s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/95a99c72ab9b479f9b2dce63839471b4 : SUCCESS in 2h 22m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4706a2b8891f4816b8d49ff34bc6dc22 : SUCCESS in 1h 01m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50c6ec4cc45748cca5a1d7e9902ba109 : SUCCESS in 55m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff69474dad23408ca7150cc03948f506 : SUCCESS in 42m 31s","accounts_in_message":[],"_revision_number":2},{"id":"d138626d4f4a7d162e54250a3f4021403ed479cb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-01 18:11:42.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/e78cf3abdeed4bcaa8c9ded2da803fe6 : POST_FAILURE in 1h 39m 42s\n- devstack-plugin-ceph-dsvm-bionic-py3 finger://zuul-v3-executor.novalocal/cad5fd0440f34a8b91891a2592e8e574 : POST_FAILURE in 2h 02m 17s","accounts_in_message":[],"_revision_number":2},{"id":"f2aff5fe8f65473a8892bf543cdf169db79a64dd","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-02-01 19:45:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"26c7a483ea4af547d6e6e138b02faf2780a834fd","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2023-02-14 19:05:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d93e8f9b54386717d2c4d86e4eabd257ff8f6ce4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 19:06:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dfa24a76ca29a54f8ed3d95d1ae894cc74396498","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 02:40:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0c4d8e76dca4b59a84d894e5b772f6f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b3e23d34cfc429e89f4841b6231dbed : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef48b4107cc6421eb57a2406d499742b : SUCCESS in 8m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51c9b6184d04444bbaf84377b787dab5 : SUCCESS in 9m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/589e10ade0454dcc8692ac578b75deda : SUCCESS in 10m 24s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/69812c9ed8d24a5f9529f0b34b903dde : SUCCESS in 1h 20m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2fe051484cb741b5b887f4c34ef967e0 : SUCCESS in 1h 01m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/546df673aaf14b19bd939fcf9145e925 : SUCCESS in 58m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e8a4f0a87a74842b2e992b293e84b50 : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":2},{"id":"b6575867ae70a824a4b23c557ead27593ea61fe1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 02:40:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1b93722f6ef8637a1e53388ad5f8587f6ae9c3ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 02:44:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09d6ef93ea6c4b729e387b4575a714ee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19dc543759ce4b8c89933754df27c50b : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"84697669d386a1841c6c59c62f0253dea23a748c","revisions":{"0ec2c754a3fb6d5cefab07bf6e67263bcdb3876d":{"kind":"REWORK","_number":1,"created":"2023-01-25 21:46:36.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/92/871792/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/871792/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/871792/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/871792/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/871792/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/871792/1"}}},"commit":{"parents":[{"commit":"6df27994d9b0ca0801a45b26c2bdf60c8eb6a268","subject":"Use the json format output of qemu-img info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6df27994d9b0ca0801a45b26c2bdf60c8eb6a268"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-01-25 21:45:48.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-01-25 21:45:48.000000000","tz":-300},"subject":"[stable-em only] move grenade jobs to experimental","message":"[stable-em only] move grenade jobs to experimental\n\nGiven that \"upgrades from Extended Maintenance are not supported\nwithin the Community\" [0], move the grenade jobs to the experimental\nqueue for the stable branches in EM status.\n\n[0] https://docs.openstack.org/project-team-guide/stable-branches.html#extended-maintenance\n\nChange-Id: Iee29e0e980faff341418a4f3c21a6ab91fe894dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0ec2c754a3fb6d5cefab07bf6e67263bcdb3876d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0ec2c754a3fb6d5cefab07bf6e67263bcdb3876d"}]},"branch":"refs/heads/stable/ussuri"},"84697669d386a1841c6c59c62f0253dea23a748c":{"kind":"REWORK","_number":2,"created":"2023-01-26 22:15:59.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/92/871792/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/871792/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/871792/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/871792/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/871792/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/871792/2"}}},"commit":{"parents":[{"commit":"6df27994d9b0ca0801a45b26c2bdf60c8eb6a268","subject":"Use the json format output of qemu-img info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6df27994d9b0ca0801a45b26c2bdf60c8eb6a268"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-01-25 21:45:48.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-01-26 21:02:59.000000000","tz":-300},"subject":"[stable-em only] move grenade jobs to experimental","message":"[stable-em only] move grenade jobs to experimental\n\nGiven that \"upgrades from Extended Maintenance are not supported\nwithin the Community\" [0], move the grenade jobs to the experimental\nqueue for the stable branches in EM status.  Additionally, remove\nthe integrated-gate-storage template because it includes grenade.\n(The other jobs in that template were already configured in check\nand gate in our .zuul.yaml, so they continue to be run.)\n\n[0] https://docs.openstack.org/project-team-guide/stable-branches.html#extended-maintenance\n\nChange-Id: Iee29e0e980faff341418a4f3c21a6ab91fe894dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/84697669d386a1841c6c59c62f0253dea23a748c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/84697669d386a1841c6c59c62f0253dea23a748c"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
