)]}'
{"id":"openstack%2Fcinder~871976","triplet_id":"openstack%2Fcinder~master~Ie8d5cb9652d3d8c7096c363e6b49ce1c75f20a23","project":"openstack/cinder","branch":"master","topic":"limit_fd_for_qemu_commands","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-02-02 13:38:15.000000000","reason":"\u003cGERRIT_ACCOUNT_29074\u003e replied on the change","reason_account":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-02-01 15:28:26.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}}},"removed_from_attention_set":{"29074":{"account":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"last_update":"2023-03-03 09:04:26.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_29074\u003e by clicking the attention icon","reason_account":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"}},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2023-01-31 18:15:11.000000000","reason":"\u003cGERRIT_ACCOUNT_9236\u003e replied on the change","reason_account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}}},"hashtags":[],"change_id":"Ie8d5cb9652d3d8c7096c363e6b49ce1c75f20a23","subject":"Limit number of file descriptors for image conversion","status":"NEW","created":"2023-01-27 10:31:41.000000000","updated":"2023-03-03 09:04:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"21a3897b569e408975a745ff3f76a71585ab0a6f","_number":871976,"virtual_id_number":871976,"owner":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-02-06 12:58:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-27 10:35:55.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-01-27 10:38:10.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-01-27 10:51:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-27 10:56:37.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-27 10:58:03.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-01-27 11:48:02.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-27 12:48:52.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-27 13:02:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-27 14:21:29.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-01-27 18:22:14.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-01-27 19:17:44.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-27 20:12:15.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-01-28 00:05:03.000000000","updated_by":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"reviewer":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"state":"CC"},{"updated":"2023-01-28 05:05:23.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-29 19:49:47.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-01-30 18:16:08.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2023-01-31 16:47:54.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-01-31 17:15:47.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2023-01-31 19:11:15.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-01-31 23:39:35.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-01 04:37:22.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-02-01 04:47:39.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2023-02-01 15:28:26.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2023-02-02 15:53:58.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-02-06 10:35:16.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-14 20:09:29.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-02-18 11:45:07.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"b066fa7591c5091353c84737612011beed292f53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2023-01-27 10:31:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"522130027d3bfb13e768866d94c3cbcc1af2ea5a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-27 10:35:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"16f42ebbbdcb3cbed75e284e21b23896139fc6ea","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-01-27 10:38:10.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/749bfd72c54144d094b8894c2f0921ad : POST_FAILURE in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"185d9462be48b8444feaa0e6620b3594d757afbe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 10:51:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db29673386f949c1b168af1ccf4fcc31\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f6339a1bf234aa4aa01b2bc88802462 : SUCCESS in 18m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e67f564ca4df46ef92634d78a7e0e411 : SUCCESS in 8m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1ef2483d6363e294cbff6c164d902afaba8fdc1","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-27 10:56:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/871976 : FAILURE in 23m 35s","accounts_in_message":[],"_revision_number":1},{"id":"6bba81c175bdc48423fcc4a93101ae76b60d4f0c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-27 10:58:03.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/76/871976/1/check/iscsi-driver-alletra-6k/32709c9/32709c9.html : FAILURE in 25m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e26fdcbc0cdb713cf8e77732397e5e3fce005990","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-27 10:59:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/871976 : FAILURE in 26m 06s","accounts_in_message":[],"_revision_number":1},{"id":"7bcc5a5ffde59a1e453f86e8045cf4fa78a4b24e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-27 11:48:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/851cadeb607a4d4a805a2d07c81d8025 : SUCCESS in 1h 14m 29s","accounts_in_message":[],"_revision_number":1},{"id":"04b6902f128c6941213977138a3893f5c83eb024","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-27 12:48:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/76/871976/1/ : FAILURE in 1h 03m 57s","accounts_in_message":[],"_revision_number":1},{"id":"db97f6999cfb6df1116d2f1e7726bf8e61e4fa96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 13:02:56.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/841b55353abd469a8909302d1f5f7ef7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/064abaa01f6849bb9c6891e3e114a5b1 : FAILURE in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b846fc8bb2f49949c9bc959de97f6f1 : SUCCESS in 10m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/765071d5fea246158c53d91f015eb51d : SUCCESS in 8m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/287b9828d5514ac9bc1bb89e95c519e4 : SUCCESS in 14m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/e6fc06593bf14dbbab93e84d3b13c09e : SUCCESS in 56m 18s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/31d70fa8a194481fab88153e9e2bdf96 : SUCCESS in 58m 59s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7b677fc47fb049bca3e86389f91cfd06 : SUCCESS in 1h 34m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8244ff20f5c44abb3e0e150dc863f90 : SUCCESS in 1h 01m 01s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/20ee45fea130486c9ad524964e9411f5 : SUCCESS in 14m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8063b58289f44e43920eed5d41ff1008 : SUCCESS in 5m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/13c2026d5cc6454fa1de89dfea7d980e : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/528398631ea34c72af9240db08e10c77 : SUCCESS in 5m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bf6e1b3375fe4197852abfddb05c20a1 : SUCCESS in 6m 21s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3692edf195ee49bdb83ecf25ac899711 : SUCCESS in 47m 38s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8199beacbe2148b6a1beab678d73da32 : FAILURE in 4m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0172bda53054471eb22051596f00dc8d : SUCCESS in 1h 12m 54s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/44af2ccd5c0f402fb51d6c5e235b56bb : FAILURE in 1h 28m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e172d197c1f149faa6393ebe3dbcfe81 : SUCCESS in 1h 59m 34s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9faffbe7a7594af8850d77c9fe7a9eb0 : SUCCESS in 1h 08m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/460281800e6e45dea2abe2cea243540f : SUCCESS in 1h 22m 56s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/aa8eedcc9ec048f4b7ffd0eed12bc228 : SUCCESS in 39m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4b2200932cc94a3e9a1942f7f1b3b763 : SUCCESS in 1h 38m 34s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8f25441997f144a6b4b878ee4f33a3cd : SUCCESS in 2h 08m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8cc91c4406d443fd822cb7f7322e4fdf : SUCCESS in 51m 39s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/380fde750b8c4820b10fe331346e0425 : SUCCESS in 1h 14m 10s","accounts_in_message":[],"_revision_number":1},{"id":"0de8307fa9acac14cc6eeb9233c0838338918b54","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-01-27 14:21:29.000000000","message":"Patch Set 1:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/1008","accounts_in_message":[],"_revision_number":1},{"id":"641b13fde804b67fa31c56992678a36b25290e08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2023-01-27 15:17:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"99106a03bb7c98a6182aa038916c67e70edd2dab","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-27 15:21:31.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"bc15d583d3027aa948d87c5469073bb0f4249a5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 15:46:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5bad8777c0f4185b06ed7943f70a83a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b45204c3a5be4c32be1f88092628b2db : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/81c989b5533d486eab5c9b73b9563505 : SUCCESS in 26m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c3575c2f1b16a246832815e7569a74a857c255a7","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-01-27 16:02:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/633e79cd2f7c416693ab9b66665c8aaf : POST_FAILURE in 6m 02s","accounts_in_message":[],"_revision_number":2},{"id":"7bee0f4bc6884b19196ab02859cb51fa26537aba","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-27 17:08:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/871976 : FAILURE in 23m 55s","accounts_in_message":[],"_revision_number":2},{"id":"2f19b7a6ed1ad2163ab3c711f70553851f139349","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-27 17:12:52.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/76/871976/2/check/iscsi-driver-alletra-6k/1387a79/1387a79.html : FAILURE in 26m 01s","accounts_in_message":[],"_revision_number":2},{"id":"b811def4210d5ffc229f7c8b20bd50fdfb9980c7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-27 17:18:56.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/871976 : FAILURE in 26m 13s","accounts_in_message":[],"_revision_number":2},{"id":"e7d43563089af6e0e7fa984b12698b007ad28835","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-27 17:33:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/41cdd1335f174772b650224e8f99a6db : SUCCESS in 1h 06m 55s","accounts_in_message":[],"_revision_number":2},{"id":"4ddfac4c331e2c5c8ed1279ce476a38ec82dec81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 17:47:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f980374543aa47669802b38ce9740724\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a812a9ca7ce348c980e0b128f9911fd8 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fba42d1a98e45ef98315532d49bec64 : SUCCESS in 10m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb5b6a00d89c443996cf1b7e65f70f36 : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/725864c5440d4a499d81ad9ad8c8a504 : SUCCESS in 13m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/f0d215356af44bd99618d3ef675430c2 : SUCCESS in 51m 32s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8413683c05d0445fbcb485abdb54a941 : SUCCESS in 58m 31s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/39bcee230f59491fb435d1ba733fd808 : SUCCESS in 1h 16m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7457e56019740609ba302de7e0085be : SUCCESS in 1h 15m 19s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b95563822ab44117b6dd3525419b33d1 : SUCCESS in 13m 33s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/59edd8aaa0804c70923fdf1209e14b58 : SUCCESS in 4m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4d96dfc561414dbbbe00a30b440e0a2f : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/54ba4f2f2fb6403da66863d619a81ddd : SUCCESS in 5m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/af1580e76c2345608876d82abae16166 : SUCCESS in 5m 07s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/14f4913448954b35a941b72a319c7551 : SUCCESS in 36m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a180d8a71f5c420cab9bb11c0e060bab : FAILURE in 4m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/dc3d55736931409da9683042b994d08a : SUCCESS in 1h 39m 19s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/cc7ae5a808514e4f96eab4dbe984ea31 : FAILURE in 1h 25m 01s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/77b918e521724457b3eab6dd0ebb648e : SUCCESS in 1h 52m 38s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/595bc21c32ae425bb698dd30c812d177 : SUCCESS in 1h 23m 51s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9a017437518f44d58ff14404e83c5303 : SUCCESS in 1h 47m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/285ddcd79dd7414f9ff3a4473c482722 : SUCCESS in 39m 10s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fab69cb0cbc04fe5bc5262dbf3b369a0 : SUCCESS in 1h 38m 33s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/192c16e6cb1845a78eb5e2c61bf2b8c0 : SUCCESS in 2h 22m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49a0684a445a420c962da811620fddc8 : SUCCESS in 49m 08s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/549244b5a8194fd98e38664313c8f561 : SUCCESS in 1h 06m 44s","accounts_in_message":[],"_revision_number":2},{"id":"765638d228bbf3c82ba5ff5e95653d2ab8f92b8c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-01-27 18:22:14.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/27_01_2023_18_14_33\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"2a729497e2614c5e0ec53de76665d15554e34784","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-27 19:17:44.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-76-871976-2/ns5_nfs : FAILURE in 1h 48m 46s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-76-871976-2/ns5_iscsi : FAILURE in 1h 52m 53s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"9aa463787dac66de39ff37bfedc9b609d9fbacb7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-27 20:12:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/871976/2/upstream-check/cinder-tempest-plugin-ontap-fcp/c196381/ : SUCCESS in 1h 32m 28s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/871976/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/f1f8752/ : SUCCESS in 44m 14s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/871976/2/upstream-check/cinder-tempest-plugin-ontap-nfs/41f28f8/ : SUCCESS in 42m 03s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/871976/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/d15f5ac/ : SUCCESS in 42m 27s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"e6db6cbf99084d21ee71102a083bedc36bc4122c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-01-27 22:28:31.000000000","message":"Patch Set 2:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/1013","accounts_in_message":[],"_revision_number":2},{"id":"2955f8c38ad8a86f6ee7558dcbb71d522ead6e6a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-27 23:56:55.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-76-871976-1/ns5_nfs : FAILURE in 13h 17m 31s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-76-871976-1/ns5_iscsi : FAILURE in 13h 4m 11s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"45bf3fdeb824fb999b224773ddfda4ec5ce7c200","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-01-28 00:05:03.000000000","message":"Patch Set 2:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/76/871976/2/fc : FAILURE in h m 0s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/76/871976/2/iscsi : FAILURE in h m 0s","accounts_in_message":[],"_revision_number":2},{"id":"3921cc5e6f34f30ef283143259fb2a76cdf87cad","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-01-28 03:12:42.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/76/871976/2/ : FAILURE in 1h 15m 18s","accounts_in_message":[],"_revision_number":2},{"id":"8bf5f375a2f9f8716296a323fd8368779e79c16d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-28 05:05:23.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/7d113bd1a56b4dd081407f4115113ce7 : RETRY_LIMIT in 1m 22s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"08d9eb0e0df0b46cfe1296907a46cbf8f43e5694","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-01-29 19:49:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/871976-2/HPMSA-iSCSI/ : SUCCESS in 1:31:22\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/871976-2/HPMSA-FC/ : SUCCESS in 1:18:04","accounts_in_message":[],"_revision_number":2},{"id":"7a0ae104afa7802aa665c3e63d40d93be267b067","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-01-30 10:40:57.000000000","message":"Patch Set 2:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/1022","accounts_in_message":[],"_revision_number":2},{"id":"b6026c7ab7a74238728c87cf5c8e5392b5681cc1","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2023-01-30 18:16:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cd30fdd85814c88e99f134d35e4201321d05cc0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2023-01-31 13:36:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"660631938c302cc2b7e1c119459db267a44cc99c","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2023-01-31 13:40:11.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"725f18d6954219665b48725e56cfa77df7561a71","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-01-31 13:40:14.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"8cd79baa3ed8c63a076e3c350dfddc4f026f8643","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-01-31 13:43:37.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/7b529a18ad75468c96f90ace4e85da90 : POST_FAILURE in 6m 14s","accounts_in_message":[],"_revision_number":3},{"id":"2fca94aa1962deca92892852c3fa7a3924a723f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 13:51:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cf6747fdd254631bb5c956fd5d8837d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/812ee1c7461342c186acb044195777ac : SUCCESS in 12m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/68a8d92f749c44628502bce3a95fa1e5 : SUCCESS in 13m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"af7bda9b0b8e63c3d155b33ea194bdc003a2ca15","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-31 14:06:25.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/871976 : FAILURE in 23m 26s","accounts_in_message":[],"_revision_number":3},{"id":"bcdf9b8428a92700493718b5e45c42dee3079774","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-01-31 14:09:47.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/871976 : FAILURE in 26m 10s","accounts_in_message":[],"_revision_number":3},{"id":"b8b3e9914a8c1ecfa8d19981b486ed734b4719e8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-01-31 14:10:36.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/76/871976/3/check/iscsi-driver-alletra-6k/ba0f642/ba0f642.html : FAILURE in 28m 46s","accounts_in_message":[],"_revision_number":3},{"id":"58c76020ef959c796676628bbd70c06be32720c4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-01-31 14:55:35.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/54f0201aa9ca4c899f9fbca7773769cc : SUCCESS in 1h 12m 39s","accounts_in_message":[],"_revision_number":3},{"id":"93bb5634d6b50ef0ba4a7250ebec8eeb30bef2c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 15:45:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d9dd599847e4171b615c7243ecc083f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85580b8926ec450ea695285551138cdf : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6adc1681d7e6431887f64d6eaa405f8c : SUCCESS in 9m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/822d4e2461774f578730c45af784a57c : SUCCESS in 10m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86397b3dbf6a464fa786759e231899e4 : SUCCESS in 12m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/a62a7b4e0fed4b28b07925284bdeba0c : SUCCESS in 58m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9e98bd15ddf8439f94ef5dfa55001fb6 : SUCCESS in 46m 22s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f770c5ef01f24744a93819e0a77bed78 : SUCCESS in 1h 19m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28232bff1b474869bfbc92cf92315242 : SUCCESS in 1h 01m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61939679b92442248498e8ce32c91c81 : SUCCESS in 8m 09s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4896771c4cf54cdca75de5e51f8d2c9a : SUCCESS in 13m 11s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/8bd03ca13884464faf2cfaeec3c37ec4 : SUCCESS in 6m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cef73e9d748849feb8940af379cadd94 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0ff8c35957714f9992754194bbb272bf : SUCCESS in 5m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/134f9793c2754ed0aa407f063e233df0 : SUCCESS in 6m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/d6a75d686d0b4b8097ada9a91f678330 : SUCCESS in 46m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8851eaf4146142b7b1ba8f09f2955dfa : FAILURE in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/82a12b770e464569a8310af60252fde2 : SUCCESS in 1h 03m 50s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/28afb033be8043989f36744a84766d1c : FAILURE in 1h 27m 36s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a937d788745045869dcf385c1fc41dc5 : SUCCESS in 1h 36m 36s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/abbef0a29d8d4157bf81314c3623c2c5 : SUCCESS in 1h 08m 08s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4cd39e12b96d44c8ac6aadfeb46c1b9f : SUCCESS in 1h 17m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/3cf80666e67f45aa9d1d732d1f96ab58 : SUCCESS in 29m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f433d4a571954cf69aefb3b96a6d5c5a : SUCCESS in 1h 42m 24s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/449e4224813c4661a0c2f4cd0d9ffb46 : SUCCESS in 2h 05m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bdcce5ef289348aaa62f25d397e4c091 : SUCCESS in 49m 47s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e84dd541db4849df836535f274200143 : SUCCESS in 1h 25m 35s","accounts_in_message":[],"_revision_number":3},{"id":"c665faf5e933e1b7ca42280c8f5eae9eb2821188","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-31 16:46:40.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/871976/3/upstream-check/cinder-tempest-plugin-ontap-fcp/03312a7/ : SUCCESS in 59m 36s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/871976/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/1c83a0d/ : FAILURE in 41m 04s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/871976/3/upstream-check/cinder-tempest-plugin-ontap-nfs/3dbffe5/ : SUCCESS in 41m 42s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/871976/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/0787b12/ : SUCCESS in 46m 08s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"4a46d6fa3751287d365e45e4e4d4d952404c1c0b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-31 16:47:54.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/76/871976/3/check/DellEMC_Unity_FC/4e7a31b/DellEMC_Unity_FC/2182 : SUCCESS in 1h 11m 16s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/76/871976/3/check/DellEMC_Unity_iSCSI/ab115a2/DellEMC_Unity_iSCSI/2049 : FAILURE in 3s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"e7109ece474b52e830b8ed08eb1582975ac470fc","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-01-31 17:15:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"46ceac04f82394802716617837c9fffb9783f9db","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2023-01-31 18:15:11.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"feeff3c219d218be3560405d9af94b45c43e6958","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-01-31 19:11:15.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/76/871976/3/check/DellEMC_PowerStore_iSCSI/e3c4b9f/DellEMC_PowerStore_iSCSI/971 : SUCCESS in 1h 16m 46s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/76/871976/3/check/DellEMC_PowerStore_FC/1f1cfa3/DellEMC_PowerStore_FC/1106 : SUCCESS in 1h 27m 44s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/76/871976/3/check/DellEMC_PowerStore_NFS/ec89e0e/DellEMC_PowerStore_NFS/610 : SUCCESS in 1h 21m 43s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/76/871976/3/check/DellEMC_PowerStore_NVMeTCP/404ae05/DellEMC_PowerStore_NVMeTCP/475 : SUCCESS in 1h 26m 53s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"03e8a5bb76fd0c638c25cbf94cfe4f7635b7d667","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-31 20:34:05.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/b9fe25ac25b1453290add35d14ca904c : POST_FAILURE in 1h 46m 44s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/b9c6065cc5c74b0f87091801027a33b3 : POST_FAILURE in 2h 18m 51s","accounts_in_message":[],"_revision_number":3},{"id":"46b931cb0e50014bf07a58d4971c8ea5c0c277f1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-01-31 23:39:35.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/76/871976/3/check/DellEMC_PowerFlex/b4216e0/DellEMC_PowerFlex/1545 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"48817b59f73a7eaa51194e19c13fd8c4f2ea3f40","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-02-01 00:05:06.000000000","message":"Patch Set 3:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/76/871976/3/fc : FAILURE in h m 0s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/76/871976/3/iscsi : FAILURE in h m 0s","accounts_in_message":[],"_revision_number":3},{"id":"6da3b9e665166d9699797603da909c2af24ed13e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-01 01:10:41.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-76-871976-3/ns5_nfs : FAILURE in 10h 58m 58s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-76-871976-3/ns5_iscsi : FAILURE in 11h 18m 8s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"3191589aeeb1c327f92cdfd9df456b13e573d9e1","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-02-01 04:37:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/871976-2/PVME-iSCSI/ : SUCCESS in 1:55:40\n- PVME-FC https://os-logs.seagate.com/dell-ci/871976-2/PVME-FC/ : SUCCESS in 1:33:56","accounts_in_message":[],"_revision_number":2},{"id":"2eb8ae26569329d44560ab28a8267765328c546e","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-02-01 04:47:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/871976-2/Seagate-iSCSI/ : SUCCESS in 1:49:47\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/871976-2/Seagate-FC/ : SUCCESS in 1:34:28","accounts_in_message":[],"_revision_number":2},{"id":"3d34e5ea5c6d287091da2f928585499b1d2b6d84","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-02-01 05:00:31.000000000","message":"Patch Set 3:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/1039","accounts_in_message":[],"_revision_number":3},{"id":"fc5443f98ae2cbb3b9d0efd18a2c8988c2904eb8","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-02-01 05:26:44.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/871976-3/HPMSA-iSCSI/ : SUCCESS in 1:34:15\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/871976-3/HPMSA-FC/ : SUCCESS in 1:15:50","accounts_in_message":[],"_revision_number":3},{"id":"1dde1a35e026b56edbc67f057f07212b107ded50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2023-02-02 13:37:32.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"5afa4de43a5341cb152d660c0b68361521447179","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2023-02-02 13:38:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"70afaa8e337b93699cda1fd1db0dc184533aeb45","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 13:47:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21c8070973db42bea1ce1249fbfc0969\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1317c2d515794708bb53b7c44229ecf0 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7d3dd4323d504871b3ea6358dd3a8b73 : SUCCESS in 8m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9d3610eb1a2dfbb5821f46101f2328b1ace93afa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-02 14:03:45.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"cb22e8fe93bb73fc1281665e567777be49202881","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-02 14:06:20.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/871976 : FAILURE in 23m 37s","accounts_in_message":[],"_revision_number":4},{"id":"c12901925b3d5b3c474960d0bfaf6981fde4a273","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-02 14:07:32.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/76/871976/4/check/iscsi-driver-alletra-6k/1aaaf92/1aaaf92.html : FAILURE in 26m 08s","accounts_in_message":[],"_revision_number":4},{"id":"c1f1aaa5a69587fb9b1b0d9b8e0d6cbd62d03858","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-02 14:09:15.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/871976 : FAILURE in 25m 59s","accounts_in_message":[],"_revision_number":4},{"id":"c29c520c39c975b4b9e2f07fc17a36525163eb4e","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-02 14:13:53.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/0160a80c69f6443bb8952a06632fed40 : POST_FAILURE in 35m 41s","accounts_in_message":[],"_revision_number":4},{"id":"60145baafe7016ffe24530e41847a18f7a871908","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-02 14:43:12.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c19677ca46a346d5bdb2ef46b64cc0f0 : SUCCESS in 1h 03m 59s","accounts_in_message":[],"_revision_number":4},{"id":"8aa18cb8e25c167815d756356a68cb4c352f1e50","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-02 15:41:59.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/76/871976/4/ : FAILURE in 1h 13m 03s","accounts_in_message":[],"_revision_number":4},{"id":"0205b73890880955e37f66b56b57b1edf75d78f7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-02 15:53:58.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/76/871976/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/55971cb/ : SUCCESS in 1h 39m 53s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/76/871976/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7ec0909/ : SUCCESS in 1h 37m 16s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/76/871976/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/f3f3881/ : SUCCESS in 1h 48m 38s","accounts_in_message":[],"_revision_number":4},{"id":"daefc09634e3f43ce36f81fed58defe4f66a1417","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 16:01:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/16785ccbf6534af9b57e4fad82441ece\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/459431d3d3834bb0b96bd84167ebc507 : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a32ecb643760426993f82dc4eba9b207 : SUCCESS in 9m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3346ed1eba5146cf857aefd4b5b3ef3b : SUCCESS in 10m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad1ef611e2a243938ecad26fc1e030a0 : SUCCESS in 12m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/eace71398248469b8b645ab0fc79280c : SUCCESS in 50m 25s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/210a07fa8bfe4aeeb38caba69641f769 : SUCCESS in 52m 51s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/80d85c59ee9d4020a1b615a440cf2694 : SUCCESS in 1h 36m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c456da0c1c34da89dc45bcc6faa1ce3 : SUCCESS in 46m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24a534251a95474189603612b0f997f7 : SUCCESS in 7m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ac6f460acecf405385ac19317140d915 : SUCCESS in 15m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/176656f38f844e24afcab3fb7c7e3f30 : SUCCESS in 4m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/020fd7fcf5aa415d88616c8cbbb42e76 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/39d579be20dc4dd39a8054df9abe119c : SUCCESS in 5m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c7e3d33eadbb440aa39d020714b178f4 : SUCCESS in 6m 12s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ca9188fd0c4a4950ae9f850ece71d17a : SUCCESS in 40m 53s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/55be3366b7d545628e607843564ff85c : FAILURE in 5m 37s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/27001a444b94476d9af1dd23e1d64eb4 : TIMED_OUT in 2h 04m 08s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2a9054c976c44f0eaadadfddca280701 : FAILURE in 1h 14m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5fe9187873e84433b132982667fd8448 : SUCCESS in 1h 56m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/bb011b0409624a7680011bc8e705a421 : FAILURE in 45m 45s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b3dcfc1d31b941eeb08a88a287d706b6 : SUCCESS in 1h 30m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d298c15b61ad461e9ed84a3705d82605 : SUCCESS in 28m 19s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c70662792e7a4e8f928f5b04a98a5e00 : SUCCESS in 1h 38m 12s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/88c0241fb5b048ccbb11f163f4a16569 : SUCCESS in 2h 16m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3147ae1028f14abbbabe552b6ab6c0c5 : SUCCESS in 41m 57s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fe686814a4e4f5fb9c19f1fd8e15181 : SUCCESS in 1h 06m 25s","accounts_in_message":[],"_revision_number":4},{"id":"15f515d023d201378c49c8515aebaa4be4dc266b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-02-02 18:21:52.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"bea8f161debcd38819232bfee74d956ad47cf51d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-02 21:22:19.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/76/871976/4/check/DellEMC_Unity_FC/0bbcf15/DellEMC_Unity_FC/2200 : SUCCESS in 1h 17m 23s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/76/871976/4/check/DellEMC_Unity_iSCSI/d9bbf4e/DellEMC_Unity_iSCSI/2069 : FAILURE in 2h 31m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"5e7781cdf6af9cfdf27575893be0fe1ad136890d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-02 22:48:22.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-76-871976-4/ns5_nfs : FAILURE in 8h 20m 23s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-76-871976-4/ns5_iscsi : FAILURE in 8h 45m 3s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"03412277572df07fafb8da3b2bdb64c5dcf0c026","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-02-03 12:25:05.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/871976-3/PVME-iSCSI/ : SUCCESS in 1:49:48\n- PVME-FC https://os-logs.seagate.com/dell-ci/871976-3/PVME-FC/ : SUCCESS in 1:33:41","accounts_in_message":[],"_revision_number":3},{"id":"6f4692bc9201273c1c4360c09a8f5618b5a06f59","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-02-03 12:25:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/871976-3/Seagate-iSCSI/ : SUCCESS in 1:52:05\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/871976-3/Seagate-FC/ : SUCCESS in 1:35:49","accounts_in_message":[],"_revision_number":3},{"id":"0ea67add6d0d88b35f61ac2610166d016d227ecc","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2023-02-06 07:30:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c70aade91f876adb1ef74adb627e9885d825ca70","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-02-06 07:33:48.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"bd6bafa49bfa7334902dce88476a6517736359e2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-06 07:46:02.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/1152c1a5a3e34a9d8388663e9197430b : FAILURE in 13m 58s","accounts_in_message":[],"_revision_number":4},{"id":"244c4c445439ca3737d8208d8caebd9b604bc6d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 07:51:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd345c7270a4430f8802bb8dbdef2b28\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67c2519a99e64aaebc4ab7d6efda580f : SUCCESS in 17m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fd17644994c544cc99c2851ed0f25695 : SUCCESS in 18m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ea29332316e96cba6846f19b2a0ad64fd50ecf55","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-06 08:02:18.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/f20b55a4a039419aa23662ff77fd9d74 : POST_FAILURE in 24m 58s","accounts_in_message":[],"_revision_number":4},{"id":"feee5b95aa7c353fb604f678e6c5fceef31ffa1d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-06 10:20:18.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/76/871976/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/fe419fe/ : FAILURE in 1h 28m 50s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/76/871976/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/cc82689/ : SUCCESS in 2h 47m 11s\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/76/871976/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/2819cfb/ : SUCCESS in 2h 04m 51s","accounts_in_message":[],"_revision_number":4},{"id":"ce00b095d8e2bc76b6e5ea18d2c8844ec6c09821","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2023-02-06 10:35:16.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/76/871976/4/check/dsvm-tempest-zadara-driver/9bc66eb : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 1h 23m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8070b906ada339758f2512a7b6919b03bd22b7b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 10:35:30.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fddb5aa581134b93a696e229d5d83c54\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8db3221696614db396a45438960d6e97 : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f75c659ae07402f9e0ee54490dafc64 : SUCCESS in 12m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf7585c9d770460db24d981fb8c0c71e : SUCCESS in 11m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9ec2e52125b43448bca184bd2dd7679 : SUCCESS in 13m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/b1918db86b99419bae07a6415e4b6f99 : SUCCESS in 1h 33m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1153b4d791c744439e129cba4c372cc8 : SUCCESS in 1h 02m 30s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b98f9ae8d43540d7a70cc5b1570daf2a : SUCCESS in 1h 41m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/940727fc3d344b5398a5ca9d1b5872a8 : SUCCESS in 1h 06m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8a4169261ae41d1bd0f4d5570f39690 : SUCCESS in 8m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c0225160893d48a3923cad88200a6062 : SUCCESS in 15m 27s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/78be00c37a6a414f84c086c318062d07 : SUCCESS in 6m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4d0875a09ba44ae9b12dbcdd871916ef : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/836ae04c840649bfbcc5f7ceb3ae07e0 : SUCCESS in 7m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c1aae953c630476d8e0b7fcdc43920ab : SUCCESS in 5m 58s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bcd3ebc6fb8c4a47942f01f3c08a21f8 : SUCCESS in 40m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2c4e3350dbe64521b81f9d8bd2670681 : FAILURE in 5m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0f3ece47108d4aa58e20501fc931b046 : SUCCESS in 1h 56m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f3279e96091340588fa35631407a025a : FAILURE in 1h 11m 25s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/706ddd63066d465c8805d60c96bcc5c9 : SUCCESS in 1h 20m 57s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/778beebb2c1f4549b93996954db5ad0e : SUCCESS in 1h 16m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bc7197c8aa514865850ebb8fff5faf55 : SUCCESS in 2h 18m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f9d0dd7abb07417b9b0bd2e90d886c38 : SUCCESS in 31m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/36b5b199fd1143d197bab303e59b5bb2 : SUCCESS in 1h 33m 42s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9b8e51450d79414e9a2246436cf6038c : TIMED_OUT in 3h 04m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73ad64ce6c9645299e30869ee644fb30 : SUCCESS in 1h 13m 11s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2724249a0cf43ecb1d7a8bdbd87ec9b : SUCCESS in 1h 21m 07s","accounts_in_message":[],"_revision_number":4},{"id":"80b3173b91c80d31a3abed7fe892c67cc5937fba","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2023-02-06 10:36:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a673e59bf75162023b6720e0d66542ef3efef711","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-06 10:38:20.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/def5cee7eef049f5adb7e2b473e884b6 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"b4a128e7444be8b6242e6db0ae02a0c03b5762aa","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-02-06 10:43:53.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"cb34747fdfd40478aa2e31c60708d1495fd8630a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 10:47:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d9ea52e2e2d419bbf1fa2eee8f34a89\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74c120e8d8cb4b8ab99f6b8891b07610 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1a1979e14e384f118fe01fd36a05b073 : SUCCESS in 9m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"83301155a1915811df847469484ab65fda6827d6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-06 11:32:02.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/69e0b61117844ed6bf883c0ad6aea800 : TIMED_OUT in 3h 16m 30s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/71cc3d0f564645e9bdb62d135549e014 : TIMED_OUT in 3h 26m 48s","accounts_in_message":[],"_revision_number":4},{"id":"1af78e5cf981b76d0954c4fe077b8ad8b2758685","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-06 11:36:15.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/e19185d708fc4f5baf094670075bb711 : POST_FAILURE in 31m 18s","accounts_in_message":[],"_revision_number":4},{"id":"3edd49d98b6cb15ae4b980ca7e89f35a9894d3bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 12:58:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/043d5082f9814564b8a31192ddb53767\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/436eb49957ae4f52b3f6091802255002 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79a65d595a774e319da9fdb72f4e8010 : SUCCESS in 11m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed7e6bb1efb74ca39248aea0e54d1945 : SUCCESS in 10m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1745effcadbf493d8f79927d1b58936e : SUCCESS in 12m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/b49e63819a59415389605866ae5c55ff : SUCCESS in 51m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/67f62e734a5d4150b4f9026489cf94c2 : SUCCESS in 1h 08m 44s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2fa0f4f9ef7240ec8baff3c331c53ce8 : SUCCESS in 1h 25m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c967ae1d627e4cea902b3a83fbf69f2d : SUCCESS in 47m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c69264e35e024f84a24fd055ebc2078e : SUCCESS in 7m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/11bc7ed4d92d462a903d9ac82fa2c013 : SUCCESS in 15m 44s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/36d9ba7033e44a2b86ec92425a7e7f05 : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a005f4ab40964792952941f87a9a3bd9 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e88fc32b72784b70907fb66f2274950d : SUCCESS in 11m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6eb71ad5abad45ccb7bb1cd722314b6e : SUCCESS in 11m 00s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/7462dcafdc3b4fa4972f4b64db4e0053 : SUCCESS in 1h 04m 57s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d1766a48939c4e5db36f960b98744626 : FAILURE in 12m 48s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9bf680fa8343443d9bd420e25bcd031b : SUCCESS in 1h 14m 37s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3d97fb6096f14b36ac98e3d500c56ed8 : FAILURE in 1h 23m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/494f4c44db2f48ed974bda49b66e3923 : SUCCESS in 1h 36m 21s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5ca3227e7c22478f87d864e4d832ae9e : SUCCESS in 1h 01m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0f83acf122dd433d96473afc71b60e1f : SUCCESS in 1h 27m 00s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/0e49706277c34b94ae0a3ad1fbfdc25d : SUCCESS in 42m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/dcb945224027447bbe562280973d3b9d : SUCCESS in 1h 28m 22s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bf784319d6ab47d39b1c68ee315dc75f : SUCCESS in 2h 01m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de7554a253c3441e8fae3030d4c10c2f : SUCCESS in 49m 55s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd76c68784da4f27a760579ce2581b8b : SUCCESS in 1h 27m 37s","accounts_in_message":[],"_revision_number":4},{"id":"fa4baaaf26e8c7a4f9d25c6a7c32e6fee770dbcb","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2023-02-06 13:28:52.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/76/871976/4/check/dsvm-tempest-zadara-driver/7bacc51 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 58m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7dcf31ee3034a16cbb8b0619a0f573cc4928c331","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-06 13:58:37.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/871976/4/upstream-check/cinder-tempest-plugin-ontap-fcp/9fd7862/ : SUCCESS in 59m 31s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/871976/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/f04daa4/ : SUCCESS in 42m 16s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/871976/4/upstream-check/cinder-tempest-plugin-ontap-nfs/aa74e86/ : SUCCESS in 46m 34s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/76/871976/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/1c43190/ : SUCCESS in 42m 55s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"f5801adfc84d03f4a6323b9eb354c29b5722f2db","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-02-06 14:37:06.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/76/871976/4/check/DellEMC_PowerStore_iSCSI/e7d6174/DellEMC_PowerStore_iSCSI/993 : SUCCESS in 1h 10m 49s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/76/871976/4/check/DellEMC_PowerStore_FC/8d2b64b/DellEMC_PowerStore_FC/1128 : SUCCESS in 1h 38m 27s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/76/871976/4/check/DellEMC_PowerStore_NFS/c368d21/DellEMC_PowerStore_NFS/632 : SUCCESS in 1h 19m 26s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/76/871976/4/check/DellEMC_PowerStore_NVMeTCP/1b1fa75/DellEMC_PowerStore_NVMeTCP/497 : SUCCESS in 1h 31m 39s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"eb99a522cc67dde08fe5385bdf7fef4be9f31906","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-02-06 14:43:22.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/76/871976/4/check/DellEMC_PowerFlex/5595dd8/DellEMC_PowerFlex/1566 : SUCCESS in 1h 44m 44s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"5f33af87d798ba6d4567f751ebc92f15aa5400f7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-06 17:47:49.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/76/871976/4/check/DellEMC_Unity_FC/ff91147/DellEMC_Unity_FC/2220 : SUCCESS in 1h 22m 30s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/76/871976/4/check/DellEMC_Unity_iSCSI/0ede68b/DellEMC_Unity_iSCSI/2089 : FAILURE in 2h 47m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"40906aa96508880fc773be393ea5b514cf8403ae","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-02-06 18:05:40.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/871976-4/Seagate-iSCSI/ : SUCCESS in 1:48:16\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/871976-4/Seagate-FC/ : SUCCESS in 1:34:03","accounts_in_message":[],"_revision_number":4},{"id":"a1ece3188b54abb291fa30d76d82f187fdb4c482","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-02-06 20:57:26.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/871976-4/HPMSA-iSCSI/ : FAILURE in 1:44:41\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/871976-4/HPMSA-FC/ : SUCCESS in 1:16:46","accounts_in_message":[],"_revision_number":4},{"id":"ddee2d03d4307be0991282d88e4228c6068bbca4","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-02-07 03:29:11.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/871976-4/PVME-iSCSI/ : SUCCESS in 1:53:21\n- PVME-FC https://os-logs.seagate.com/dell-ci/871976-4/PVME-FC/ : SUCCESS in 1:34:27","accounts_in_message":[],"_revision_number":4},{"id":"503f422e6ca8b9cf005720126dac575ec9fbf33a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-02-14 20:09:29.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/76/871976/4/20230215020804/NEC-V-FC : SUCCESS in 2h 01m 29s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/76/871976/4/20230215020804/NEC-V-ISCSI : SUCCESS in 3h 01m 22s","accounts_in_message":[],"_revision_number":4},{"id":"ac71154d215ec4f75721afc3a3b1aeb9d9aa5ec3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-02-18 11:45:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/76/871976/4/20230218171248/NEC-FC : SUCCESS in 3h 32m 09s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/76/871976/4/20230218171248/NEC-ISCSI : SUCCESS in 3h 32m 16s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"16ded6eab19f02b2c25985a49f4792c796c4b579","revisions":{"20aab056fb5fab4e15b68c9657fa4d81a2261d27":{"kind":"REWORK","_number":1,"created":"2023-01-27 10:31:41.000000000","uploader":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"ref":"refs/changes/76/871976/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/76/871976/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/871976/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/871976/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/871976/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/76/871976/1"}}},"commit":{"parents":[{"commit":"7d250602437978afc27725d6c93e11d55b72fab2","subject":"Merge \"Send the correct location URI to the Glance v2 API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d250602437978afc27725d6c93e11d55b72fab2"}]}],"author":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2023-01-26 14:14:11.000000000","tz":60},"committer":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2023-01-27 09:44:29.000000000","tz":60},"subject":"Limit number of file for qemu calls","message":"Limit number of file for qemu calls\n\nThis is a followup of [1] handling a potential denial of service\nvulnerbility in cinder-volume.\nThe vulnerability is currently not exploitable (see description below).\n\nIn case you have a vmdk file like the following:\n```\nversion\u003d1\nCID\u003db1a17f47\nparentCID\u003dffffffff\ncreateType\u003d\"monolithicFlat\"\n\nRW 1 FLAT \"/etc/passwd\" 0\nRW 1 FLAT \"/etc/passwd\" 0\n[... assume here are now 1000000 lines as above ...]\nRW 1 FLAT \"/etc/passwd\" 0\nRW 1 FLAT \"/etc/passwd\" 0\n\nddb.virtualHWVersion \u003d \"4\"\nddb.geometry.cylinders \u003d \"20805\"\nddb.geometry.heads \u003d \"16\"\nddb.geometry.sectors \u003d \"63\"\nddb.adapterType \u003d \"ide\"\nddb.toolsVersion \u003d \"2147483647\"\n```\n\nIn this case the initial `qemu-img info` call in\n`_create_from_image_cache_or_download` will take a long time to complete\nand consume a large amount of memory. `qemu-img` will be terminated by\nthe `image_conversion_cpu_limit` after 60 seconds per default. It will\nhave consumed a full cpu core for the entire time of it running.\n\nIf you now assume a user can create a large amount of volumes in\nparallel from such a malicious image, the user would cause a\ndenial-of-service of the node cinder-volume runs on by exhausing the cpu\ncores available there.\n\nAs `qemu-img info` will only close the files after all lines have been\nprocesses we can limit the processing by setting a maximum number of\nopen files.\n\nI choose 1024 here as this is already accidentially enforced by default\nand we can be quite sure that no reasonable image will ever need 1024 files\nopen at once during conversion. If this is set then the `qemu-img info` call\nterminates after 0.5 seconds. This siginificantly reduces the denial-of-service\npotential.\n\nWhy this is currently not exploitable:\n1. The call in `_create_from_image_cache_or_download` for `qemu_img_info`\n   is using the default `run_as_root\u003dTrue`\n2. `qemu_img_info` specifies a `prlimit` for address space and cpu time\n3. The processes being executed then are (in order)\n   1. `oslo_concurrency.prlimit` calling\n   2. `sudo` calling\n   3. `cinder-rootwrap` calling\n   4. `qemu-img`\n4. the resource limits applied are:\n   1. for `oslo_concurrency.prlimit`: the limits for address space and\n      cpu time\n   2. for `cinder-rootwrap`: a limit for the amount of file descriptors\n      (to 1024 per default) at [2]\n5. the \"accidential\" limit by `cinder-rootwrap` here (see comments at [2])\n   prevents exploitation. However from the comments of `olso.rootwrap` this\n   does not look like an actual security feature, but rather a fix for the\n   shortcomings of python2. Therefor we should still set the value\n   explicitly on the cinder side to prevent exploitation once the code\n   of `oslo.rootwrap` might be removed in the future.\n\n[1]: https://bugs.launchpad.net/nova/+bug/1996188\n[2]: https://opendev.org/openstack/oslo.rootwrap/src/commit/349603af2ea0d2da21adc50b317936c58480c202/oslo_rootwrap/cmd.py#L91-L118\n\nRelated-Bug: #1996188\nChange-Id: Ie8d5cb9652d3d8c7096c363e6b49ce1c75f20a23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/20aab056fb5fab4e15b68c9657fa4d81a2261d27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/20aab056fb5fab4e15b68c9657fa4d81a2261d27"}]},"branch":"refs/heads/master"},"7e84186385d14942b82c4a7fdad6f1e24de0c18f":{"kind":"REWORK","_number":2,"created":"2023-01-27 15:17:08.000000000","uploader":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"ref":"refs/changes/76/871976/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/76/871976/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/871976/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/871976/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/871976/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/76/871976/2"}}},"commit":{"parents":[{"commit":"7d250602437978afc27725d6c93e11d55b72fab2","subject":"Merge \"Send the correct location URI to the Glance v2 API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d250602437978afc27725d6c93e11d55b72fab2"}]}],"author":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2023-01-26 14:14:11.000000000","tz":60},"committer":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2023-01-27 15:16:55.000000000","tz":60},"subject":"Limit number of file for qemu calls","message":"Limit number of file for qemu calls\n\nThis is a followup of [1] handling a potential denial of service\nvulnerbility in cinder-volume.\nThe vulnerability is currently not exploitable (see description below).\n\nIn case you have a vmdk file like the following:\n```\nversion\u003d1\nCID\u003db1a17f47\nparentCID\u003dffffffff\ncreateType\u003d\"monolithicFlat\"\n\nRW 1 FLAT \"/etc/passwd\" 0\nRW 1 FLAT \"/etc/passwd\" 0\n[... assume here are now 1000000 lines as above ...]\nRW 1 FLAT \"/etc/passwd\" 0\nRW 1 FLAT \"/etc/passwd\" 0\n\nddb.virtualHWVersion \u003d \"4\"\nddb.geometry.cylinders \u003d \"20805\"\nddb.geometry.heads \u003d \"16\"\nddb.geometry.sectors \u003d \"63\"\nddb.adapterType \u003d \"ide\"\nddb.toolsVersion \u003d \"2147483647\"\n```\n\nIn this case the initial `qemu-img info` call in\n`_create_from_image_cache_or_download` will take a long time to complete\nand consume a large amount of memory. `qemu-img` will be terminated by\nthe `image_conversion_cpu_limit` after 60 seconds per default. It will\nhave consumed a full cpu core for the entire time of it running.\n\nIf you now assume a user can create a large amount of volumes in\nparallel from such a malicious image, the user would cause a\ndenial-of-service of the node cinder-volume runs on by exhausing the cpu\ncores available there.\n\nAs `qemu-img info` will only close the files after all lines have been\nprocesses we can limit the processing by setting a maximum number of\nopen files.\n\nI choose 1024 here as this is already accidentially enforced by default\nand we can be quite sure that no reasonable image will ever need 1024 files\nopen at once during conversion. If this is set then the `qemu-img info` call\nterminates after 0.5 seconds. This siginificantly reduces the denial-of-service\npotential.\n\nWhy this is currently not exploitable:\n1. The call in `_create_from_image_cache_or_download` for `qemu_img_info`\n   is using the default `run_as_root\u003dTrue`\n2. `qemu_img_info` specifies a `prlimit` for address space and cpu time\n3. The processes being executed then are (in order)\n   1. `oslo_concurrency.prlimit` calling\n   2. `sudo` calling\n   3. `cinder-rootwrap` calling\n   4. `qemu-img`\n4. the resource limits applied are:\n   1. for `oslo_concurrency.prlimit`: the limits for address space and\n      cpu time\n   2. for `cinder-rootwrap`: a limit for the amount of file descriptors\n      (to 1024 per default) at [2]\n5. the \"accidential\" limit by `cinder-rootwrap` here (see comments at [2])\n   prevents exploitation. However from the comments of `olso.rootwrap` this\n   does not look like an actual security feature, but rather a fix for the\n   shortcomings of python2. Therefor we should still set the value\n   explicitly on the cinder side to prevent exploitation once the code\n   of `oslo.rootwrap` might be removed in the future.\n\n[1]: https://bugs.launchpad.net/nova/+bug/1996188\n[2]: https://opendev.org/openstack/oslo.rootwrap/src/commit/349603af2ea0d2da21adc50b317936c58480c202/oslo_rootwrap/cmd.py#L91-L118\n\nRelated-Bug: #1996188\nChange-Id: Ie8d5cb9652d3d8c7096c363e6b49ce1c75f20a23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e84186385d14942b82c4a7fdad6f1e24de0c18f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7e84186385d14942b82c4a7fdad6f1e24de0c18f"}]},"branch":"refs/heads/master"},"a41ae92aeffed85720dd865edecbcfee9bee31af":{"kind":"REWORK","_number":3,"created":"2023-01-31 13:36:44.000000000","uploader":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"ref":"refs/changes/76/871976/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/76/871976/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/871976/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/871976/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/871976/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/76/871976/3"}}},"commit":{"parents":[{"commit":"7d250602437978afc27725d6c93e11d55b72fab2","subject":"Merge \"Send the correct location URI to the Glance v2 API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d250602437978afc27725d6c93e11d55b72fab2"}]}],"author":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2023-01-26 14:14:11.000000000","tz":60},"committer":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2023-01-31 13:36:36.000000000","tz":60},"subject":"Limit number of file descriptors for image conversion","message":"Limit number of file descriptors for image conversion\n\nThis is a followup of [1] handling a potential denial of service\nvulnerbility in cinder-volume.\nThe vulnerability is currently not exploitable (see description below).\n\nIn case you have a vmdk file like the following:\n```\nversion\u003d1\nCID\u003db1a17f47\nparentCID\u003dffffffff\ncreateType\u003d\"monolithicFlat\"\n\nRW 1 FLAT \"/etc/passwd\" 0\nRW 1 FLAT \"/etc/passwd\" 0\n[... assume here are now 1000000 lines as above ...]\nRW 1 FLAT \"/etc/passwd\" 0\nRW 1 FLAT \"/etc/passwd\" 0\n\nddb.virtualHWVersion \u003d \"4\"\nddb.geometry.cylinders \u003d \"20805\"\nddb.geometry.heads \u003d \"16\"\nddb.geometry.sectors \u003d \"63\"\nddb.adapterType \u003d \"ide\"\nddb.toolsVersion \u003d \"2147483647\"\n```\n\nIn this case the initial `qemu-img info` call in\n`_create_from_image_cache_or_download` will take a long time to complete\nand consume a large amount of memory. `qemu-img` will be terminated by\nthe `image_conversion_cpu_limit` after 60 seconds per default. It will\nhave consumed a full cpu core for the entire time of it running.\n\nIf you now assume a user can create a large amount of volumes in\nparallel from such a malicious image, the user would cause a\ndenial-of-service of the node cinder-volume runs on by exhausing the cpu\ncores available there.\n\nAs `qemu-img info` will only close the files after all lines have been\nprocesses we can limit the processing by setting a maximum number of\nopen files.\n\nI choose 1024 here as this is already accidentially enforced by default\nand we can be quite sure that no reasonable image will ever need 1024 files\nopen at once during conversion. If this is set then the `qemu-img info` call\nterminates after 0.5 seconds. This siginificantly reduces the denial-of-service\npotential.\n\nWhy this is currently not exploitable:\n1. The call in `_create_from_image_cache_or_download` for `qemu_img_info`\n   is using the default `run_as_root\u003dTrue`\n2. `qemu_img_info` specifies a `prlimit` for address space and cpu time\n3. The processes being executed then are (in order)\n   1. `oslo_concurrency.prlimit` calling\n   2. `sudo` calling\n   3. `cinder-rootwrap` calling\n   4. `qemu-img`\n4. the resource limits applied are:\n   1. for `oslo_concurrency.prlimit`: the limits for address space and\n      cpu time\n   2. for `cinder-rootwrap`: a limit for the amount of file descriptors\n      (to 1024 per default) at [2]\n5. the \"accidential\" limit by `cinder-rootwrap` here (see comments at [2])\n   prevents exploitation. However from the comments of `olso.rootwrap` this\n   does not look like an actual security feature, but rather a fix for the\n   shortcomings of python2. Therefor we should still set the value\n   explicitly on the cinder side to prevent exploitation once the code\n   of `oslo.rootwrap` might be removed in the future.\n\n[1]: https://bugs.launchpad.net/nova/+bug/1996188\n[2]: https://opendev.org/openstack/oslo.rootwrap/src/commit/349603af2ea0d2da21adc50b317936c58480c202/oslo_rootwrap/cmd.py#L91-L118\n\nRelated-Bug: #1996188\nChange-Id: Ie8d5cb9652d3d8c7096c363e6b49ce1c75f20a23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a41ae92aeffed85720dd865edecbcfee9bee31af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a41ae92aeffed85720dd865edecbcfee9bee31af"}]},"branch":"refs/heads/master"},"16ded6eab19f02b2c25985a49f4792c796c4b579":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-02-02 13:37:32.000000000","uploader":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"ref":"refs/changes/76/871976/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/76/871976/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/871976/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/871976/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/76/871976/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/76/871976/4"}}},"commit":{"parents":[{"commit":"7d250602437978afc27725d6c93e11d55b72fab2","subject":"Merge \"Send the correct location URI to the Glance v2 API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7d250602437978afc27725d6c93e11d55b72fab2"}]}],"author":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2023-01-26 14:14:11.000000000","tz":60},"committer":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2023-02-02 13:36:37.000000000","tz":60},"subject":"Limit number of file descriptors for image conversion","message":"Limit number of file descriptors for image conversion\n\nThis is a followup of [1] handling a potential denial of service\nvulnerbility in cinder-volume.\nThe vulnerability is currently not exploitable (see description below).\n\nIn case you have a vmdk file like the following:\n```\nversion\u003d1\nCID\u003db1a17f47\nparentCID\u003dffffffff\ncreateType\u003d\"monolithicFlat\"\n\nRW 1 FLAT \"/etc/passwd\" 0\nRW 1 FLAT \"/etc/passwd\" 0\n[... assume here are now 1000000 lines as above ...]\nRW 1 FLAT \"/etc/passwd\" 0\nRW 1 FLAT \"/etc/passwd\" 0\n\nddb.virtualHWVersion \u003d \"4\"\nddb.geometry.cylinders \u003d \"20805\"\nddb.geometry.heads \u003d \"16\"\nddb.geometry.sectors \u003d \"63\"\nddb.adapterType \u003d \"ide\"\nddb.toolsVersion \u003d \"2147483647\"\n```\n\nIn this case the initial `qemu-img info` call in\n`_create_from_image_cache_or_download` will take a long time to complete\nand consume a large amount of memory. `qemu-img` will be terminated by\nthe `image_conversion_cpu_limit` after 60 seconds per default. It will\nhave consumed a full cpu core for the entire time of it running.\n\nIf you now assume a user can create a large amount of volumes in\nparallel from such a malicious image, the user would cause a\ndenial-of-service of the node cinder-volume runs on by exhausing the cpu\ncores available there.\n\nAs `qemu-img info` will only close the files after all lines have been\nprocesses we can limit the processing by setting a maximum number of\nopen files.\n\nI choose 1024 here as this is already accidentially enforced by default\nand we can be quite sure that no reasonable image will ever need 1024 files\nopen at once during conversion. If this is set then the `qemu-img info` call\nterminates after 0.5 seconds. This siginificantly reduces the denial-of-service\npotential.\n\nWhy this is currently not exploitable:\n1. The call in `_create_from_image_cache_or_download` for `qemu_img_info`\n   is using the default `run_as_root\u003dTrue`\n2. `qemu_img_info` specifies a `prlimit` for address space and cpu time\n3. The processes being executed then are (in order)\n   1. `oslo_concurrency.prlimit` calling\n   2. `sudo` calling\n   3. `cinder-rootwrap` calling\n   4. `qemu-img`\n4. the resource limits applied are:\n   1. for `oslo_concurrency.prlimit`: the limits for address space and\n      cpu time\n   2. for `cinder-rootwrap`: a limit for the amount of file descriptors\n      (to 1024 per default) at [2]\n5. the \"accidential\" limit by `cinder-rootwrap` here (see comments at [2])\n   prevents exploitation. However from the comments of `olso.rootwrap` this\n   does not look like an actual security feature, but rather a fix for the\n   shortcomings of python2. Therefor we should still set the value\n   explicitly on the cinder side to prevent exploitation once the code\n   of `oslo.rootwrap` might be removed in the future.\n\n[1]: https://bugs.launchpad.net/nova/+bug/1996188\n[2]: https://opendev.org/openstack/oslo.rootwrap/src/commit/349603af2ea0d2da21adc50b317936c58480c202/oslo_rootwrap/cmd.py#L91-L118\n\nCloses-Bug: #2004565\nChange-Id: Ie8d5cb9652d3d8c7096c363e6b49ce1c75f20a23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/16ded6eab19f02b2c25985a49f4792c796c4b579"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/16ded6eab19f02b2c25985a49f4792c796c4b579"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
