)]}'
{"id":"openstack%2Fcinder~872307","triplet_id":"openstack%2Fcinder~stable%2Fzed~I3256875a33a30560b834b4c8a0c6bb1b5edf69fa","project":"openstack/cinder","branch":"stable/zed","topic":"latest_failover_fix-stable/zed","attention_set":{},"removed_from_attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2023-04-10 17:04:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3256875a33a30560b834b4c8a0c6bb1b5edf69fa","subject":"[Pure Storage] Fix issue with loss of replicated array","status":"MERGED","created":"2023-02-01 15:44:04.000000000","updated":"2023-04-10 17:06:16.000000000","submitted":"2023-04-10 17:04:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"872307-latest_failover_fix-stable/zed","cherry_pick_of_change":855060,"cherry_pick_of_patch_set":13,"meta_rev_id":"1d3a764c868d1523c53e7e9f6f9518aa3b90889a","_number":872307,"virtual_id_number":872307,"owner":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-10 17:04:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":2,"date":"2023-04-10 15:50:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-03-23 14:19:40.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":2,"date":"2023-02-28 10:36:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":1,"date":"2023-04-10 15:50:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-01 15:45:47.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-02-01 15:46:16.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-01 15:47:11.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-02-01 15:57:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-01 16:12:43.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-01 17:02:56.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-01 18:58:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-01 21:03:34.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-02-28 10:36:21.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-03-23 14:19:40.000000000","updated_by":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2023-04-10 15:50:17.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"}],"messages":[{"id":"207466825b83a042ff0c542d06fe91631034786a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-02-01 15:44:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ea8eef666ad60bd9b866018a1c40bec845c222b7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-01 15:45:47.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"b7d4fcb0d60ee62bf20547e52e80fa90600125a2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-01 15:46:16.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-07-872307-1/ns5_nfs : FAILURE in 1m 38s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-07-872307-1/ns5_iscsi : FAILURE in 1m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"c2837b17b718414e286733b7e1790868ee736021","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-02-01 15:47:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"c79c1429632f31c3607e8f1f2646facb008d4bea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 15:57:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a778ddeecc0b4978a5104c82637cc572\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f368c7de3ee4165bc0d02e8eb70005a : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f21b2d3f658e44ff9992ce9b105fe2a1 : SUCCESS in 11m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd0264ae0e394d5cb2e795b7f4c204b02c49cb2c","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-01 16:12:43.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/872307 : FAILURE in 26m 20s","accounts_in_message":[],"_revision_number":1},{"id":"7a6425b66794f92c0df3bbea2264177af4c15957","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-01 16:41:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/872307 : FAILURE in 55m 23s","accounts_in_message":[],"_revision_number":1},{"id":"49e5cf54506e528053eedcb7daa12becee9fd1a2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-01 16:43:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/07/872307/1/check/iscsi-driver-alletra-6k/8688762/8688762.html : FAILURE in 56m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0ca3ba511cad12aca177eda6f6d2a03e88abdc0a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-01 17:02:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/07/872307/1/ : FAILURE in 1h 10m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a42d9932175f9a7b917572f3d614721888d70522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 18:58:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/445d1076e73a41978966db80adad4f30\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f28f167b4d3b4f19805f0df9200df63d : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ef61fc29822401eaaa30d09070a8412 : SUCCESS in 9m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa9f1843915f415aaa6c4dd171536ccd : SUCCESS in 10m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e18f32cf87c4b5abcf034821e4d27da : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef291f6f7dca48c7bdd5e0d96fbfe405 : SUCCESS in 12m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/dbd128823f404ebd820422670a8f4036 : SUCCESS in 50m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de406cae7165447384629571dc33b4e4 : SUCCESS in 1h 34m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef1fa72ca46540e19577ab233001b5e8 : SUCCESS in 1h 01m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/957c941a4c3b4239aa27e353490d06cb : SUCCESS in 7m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/dff82d77e4de49bfbbf55ecab1b74e9a : SUCCESS in 17m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/04fb10ebf1bb4efabecb1ed59a2aedcf : SUCCESS in 6m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/782b17b5262140709c3002fcdd0538e0 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c41037db89664cf1a51f669d5e04ffb9 : SUCCESS in 5m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a64957a04ec040c19dabbbbeaeede5c9 : SUCCESS in 5m 19s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1cca94f3170045babbc8bf5729c25ea9 : SUCCESS in 45m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5ade7cab1b204e44aad3ebb6d35fdaf6 : FAILURE in 5m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3e5be40479e5465892fca6050d439b94 : SUCCESS in 1h 50m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/96f915ab6af04a53b2c1991c2c035397 : FAILURE in 1h 23m 35s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/802254966e99463ab5be48c65fd4de08 : SUCCESS in 1h 59m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/48b195b8a53a433abb87af8989862e00 : FAILURE in 1h 16m 22s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a6857c2747294508a8694f9e1d172b96 : SUCCESS in 1h 37m 41s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/20e3d179866f49e1a1bd9ab043180c9d : SUCCESS in 51m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/a484b1e82c434f5483489ddfd5d5bacb : SUCCESS in 1h 39m 12s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6208379576a74d16a420bdb22fb68f29 : SUCCESS in 2h 11m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f6eb842ff43455e960555e982674872 : SUCCESS in 54m 26s","accounts_in_message":[],"_revision_number":1},{"id":"cfaf0f40f37449633bda98ae62c01d77dc16daa4","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-02-01 19:10:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8455ad2cc51c17ccac4cebe3eb48f3fa6483eb2e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-02-01 19:11:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"efd18ebbd02dceb32ed51f9a1037647cf23aa141","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 19:19:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4537d863cd742a69291e75e34efa6b4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b295c57d371b4f0f8c28eba9b28233ed : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d37ffa6ddcc04d4580b26c14fec0cb5f : SUCCESS in 8m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b90176f71d8377a1d544f02b2bdc19e64677da0e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-01 21:03:34.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/07/872307/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/75cf7a7/ : SUCCESS in 1h 51m 06s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/07/872307/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/0ad54b1/ : SUCCESS in 1h 35m 58s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/07/872307/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/471c718/ : SUCCESS in 1h 38m 28s","accounts_in_message":[],"_revision_number":1},{"id":"8d2c6a817873346a9847b27fc818b7f79a84d518","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 21:04:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/825c531999c54d21b5666073d0bde60f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f04c208020f4cc7a2111ea7be47221f : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48af3f0bd21f44d4906ac4e67c7b8d12 : SUCCESS in 10m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39596e2939654d52a8fa3a950edd5dce : SUCCESS in 10m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1a7b1fc017448819b8aeb2712df8500 : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa0ba1d305934607bce24e22d3f3a98d : SUCCESS in 11m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/0c860ee963fa4f4b96a9b238ec63bc68 : SUCCESS in 56m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/02e0f7e22b404ca78e315d1e2cafbb14 : SUCCESS in 1h 35m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2bb734e3b384a7f9d465ca8b9317d04 : SUCCESS in 53m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38f6f0d957f1469da272b23a124c39e5 : SUCCESS in 9m 11s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/08a95ab2a6a0408e89792c032dc4f38b : SUCCESS in 14m 02s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/77f9c0fedb1f49a288fec58cd5b8df7f : SUCCESS in 5m 02s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9f5f8223c6c8423fb5e46703a16f26fb : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/93dcc38285fc48aca3a5ede32794eb7c : SUCCESS in 5m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4931236dd04c47d08dfb394dae032855 : SUCCESS in 5m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9063169e6a264917a83c904e6c678598 : SUCCESS in 39m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/38b1054af09e49799f4d4b435a374f54 : FAILURE in 5m 44s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0ccb62b58d27439f8dbe6c919ca60923 : SUCCESS in 1h 11m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/682450f7ee2b48d592e5e81de2a0ccf7 : FAILURE in 1h 19m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7ed63546f3ad4ceea6a2cb4ee9f2f21d : SUCCESS in 1h 30m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/c534c27e164a4f9987c21318b95cbd24 : SUCCESS in 1h 04m 36s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/33e319bd87724b93ab5bdc1cb1538b37 : SUCCESS in 1h 21m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e391dfb3c2084784838b83ee2ab07492 : SUCCESS in 36m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/809d2d7003674b5f92ec8453febd7534 : SUCCESS in 1h 15m 43s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6e10a8b5430d4103a1fe321ab0e69c80 : SUCCESS in 1h 50m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28030bf13a9146a89c885bd798fbe2a3 : SUCCESS in 48m 38s","accounts_in_message":[],"_revision_number":1},{"id":"149afd3ea7142cff45a7f3e50648667576b9620a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-02-28 10:36:21.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df16ec67009d22cd9eb983eb89360b4653603b0e","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2023-03-23 14:19:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb7b71e8f53378d84490877888759ca8731a4583","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-04-10 15:50:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"86a3e5064c8968f32307b32de744d27bf4ac0705","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 15:50:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"19fda3c42691e7481971fd2219c5da5e1f3cefc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 17:04:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07f754963ce2499bbd4af96be7c3c1d4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/419b361a794f4ef295ec9b585898fc94 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d10b01048654c6b97cfdbf19e8f59f5 : SUCCESS in 8m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bde6368f19274d31ba43ca4a64fa4ebe : SUCCESS in 9m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3700b7d58094147b7ed6e25da1586d3 : SUCCESS in 12m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/0aec4d69885c48d485fde3f22092aa80 : SUCCESS in 44m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2897398fee904d85aacff36ea8700d3a : SUCCESS in 55m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82fa738ecfdc4de2ad9148242fa6c82f : SUCCESS in 41m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b240d17a832c4364b1a2d8d6fc2483b4 : SUCCESS in 8m 55s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/5f94fc89c14e4f40aefea7c2da061602 : SUCCESS in 1h 09m 04s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c85ddc83140b46a18fe165ffbef681ac : SUCCESS in 52m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ec5bebd136244498d0bc3d9ed93f318 : SUCCESS in 59m 04s","accounts_in_message":[],"_revision_number":1},{"id":"86312f144c8cb79c4fcdac5b0c511209d21ba689","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 17:04:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1d3a764c868d1523c53e7e9f6f9518aa3b90889a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 17:06:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9f10b21a8ea40649f25433f7c2f42e6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7a01ec5ea5e402093c06b7b0ddf6a8d : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41972b5b1d734a9685dce3fa3b855f13 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9d12d263e51c0f1b38a51571aa624862e229f203","revisions":{"9d12d263e51c0f1b38a51571aa624862e229f203":{"kind":"REWORK","_number":1,"created":"2023-02-01 15:44:04.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/07/872307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/07/872307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/872307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/872307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/07/872307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/07/872307/1"}}},"commit":{"parents":[{"commit":"c9459d9ac366f7a685a61bd9739b3126749961b9","subject":"Merge \"Check VMDK subformat against an allowed list\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c9459d9ac366f7a685a61bd9739b3126749961b9"}]}],"author":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2022-05-31 18:51:51.000000000","tz":-240},"committer":{"name":"Simon Dodsley","email":"simon@purestorage.com","date":"2023-02-01 15:44:04.000000000","tz":0},"subject":"[Pure Storage] Fix issue with loss of replicated array","message":"[Pure Storage] Fix issue with loss of replicated array\n\nWhen an array on either side of a replicated configuration loses\nconnectivity to Cinder this can cause the driver to stop working.\n\nThis patch resolves that issue by validating all arrays in a\nreplicated pair exist and raising a warning if that is not\nthe case.\n\nCloses Bug: #1969784\nCo-Authored-By: Keerthivasan \u003cksuresh@purestorage.com\u003e\n\nChange-Id: I3256875a33a30560b834b4c8a0c6bb1b5edf69fa\n(cherry picked from commit deeee2b363c4f6acf0e5cfaa6b2b04b6eb6c8362)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d12d263e51c0f1b38a51571aa624862e229f203"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9d12d263e51c0f1b38a51571aa624862e229f203"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
