)]}'
{"id":"openstack%2Fcinder~873862","triplet_id":"openstack%2Fcinder~master~I99edfa2b18f3315d7ff81981216dccb8e32e0fe4","project":"openstack/cinder","branch":"master","topic":"873862","attention_set":{"33436":{"account":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"last_update":"2023-07-18 04:00:09.000000000","reason":"Someone else replied on the change"},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2023-07-17 05:52:13.000000000","reason":"\u003cGERRIT_ACCOUNT_33436\u003e replied on the change","reason_account":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"}}},"removed_from_attention_set":{"30615":{"account":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"last_update":"2023-05-24 05:54:03.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},"35067":{"account":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"last_update":"2023-05-24 05:54:03.000000000","reason":"\u003cGERRIT_ACCOUNT_29122\u003e replied on the change","reason_account":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}}},"hashtags":[],"change_id":"I99edfa2b18f3315d7ff81981216dccb8e32e0fe4","subject":"cinder-backup:use snapshot_id create backup","status":"NEW","created":"2023-02-17 08:53:33.000000000","updated":"2023-07-18 04:00:09.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"053d243ff3ad4d87679b5161a55a75c8d3cbc2e6","_number":873862,"virtual_id_number":873862,"owner":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-23 12:33:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"all":[{"value":1,"date":"2023-05-24 05:54:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}],"REVIEWER":[{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-17 09:03:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-17 09:19:31.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-17 09:20:29.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-02-17 09:47:46.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-17 09:57:23.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-17 10:03:59.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-02-17 10:33:53.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-17 11:07:17.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-02-17 11:21:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-17 11:25:34.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-02-17 11:29:29.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-02-17 11:35:41.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2023-02-17 12:23:10.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-02-17 12:48:42.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-17 12:50:10.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-02-17 13:05:04.000000000","updated_by":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"reviewer":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"state":"CC"},{"updated":"2023-02-17 14:38:15.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-02-17 18:20:37.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-18 03:09:25.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-18 23:31:46.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-02-19 17:56:05.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2023-02-19 18:14:15.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-02-21 18:47:18.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-22 12:28:04.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2023-02-23 08:15:12.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-02-28 02:40:44.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-16 10:13:33.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-05-16 11:34:26.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-05-17 06:04:42.000000000","updated_by":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"reviewer":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"state":"REVIEWER"},{"updated":"2023-05-24 02:47:26.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-24 05:54:03.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2023-07-17 05:52:13.000000000","updated_by":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2023-07-18 04:00:09.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"}],"messages":[{"id":"c634b98593497a447daf62869e623c0a14bca637","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"date":"2023-02-17 08:53:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38c6217178c4f82530513eb645fe90fe1b00990c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 09:03:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69c5648245e244c1a2496fa31b00092c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/908d52c33fdd46cb9a0aaa649c19ef41 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/56d15ab7527f475290b1097e1a4aa37d : SUCCESS in 8m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1cf8c91996e3d43b2b1385d34c40fb3caeb7d4b6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-17 09:19:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/873862 : FAILURE in 23m 33s","accounts_in_message":[],"_revision_number":1},{"id":"5db89831d6e68b691f4b7785ad0c7a3f083e9c8a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-17 09:20:29.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/62/873862/1/check/iscsi-driver-alletra-6k/7c2156b/7c2156b.html : FAILURE in 25m 17s","accounts_in_message":[],"_revision_number":1},{"id":"274578ba98c40c2cc23df305edd9e334e7a8c775","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-17 09:21:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/873862 : FAILURE in 26m 21s","accounts_in_message":[],"_revision_number":1},{"id":"9e6b7c424e5161000530c9e52b5eacedf3bc3e80","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2023-02-17 09:47:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver http://openstack-ci-logs.zadarastorage.com/62/873862/1/check/dsvm-tempest-zadara-driver/8c383da : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 53m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e89e03e785d36bd30800714b6a61a65d9c6c827e","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-17 09:57:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/62/873862/1/ : FAILURE in 56m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d2a5a2a4d8236b111890285ecd6785a5ff22b592","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-17 10:03:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/9f82e9791aa247d8ba5f5c054dfb868b : SUCCESS in 1h 09m 41s","accounts_in_message":[],"_revision_number":1},{"id":"81c64cb88473f618743c2fb10850217512a87eec","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-02-17 10:33:53.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/62/873862/1/check/fujitsu-eternus-dx-iscsi/637ce84 : SUCCESS in 1h 32m 52s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/62/873862/1/check/fujitsu-eternus-dx-fc/f6974de : SUCCESS in 1h 39m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d90b34dd1fca226514ac62871254ee0f7ea48c89","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-17 11:07:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/62/873862/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/1d11898/ : SUCCESS in 1h 37m 27s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/62/873862/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a67f020/ : FAILURE in 2h 11m 04s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/62/873862/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/8378d79/ : SUCCESS in 1h 36m 17s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/62/873862/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/b117870/ : SUCCESS in 1h 45m 57s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/62/873862/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/59eb363/ : SUCCESS in 2h 01m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c49fc5475a4c1c6879ff9b1d0622140cf9658a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 11:21:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b24a71bda684c2ea3d5d690116d8525\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25be9416e2c340119b9de9bbc653a5c1 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6431fec57da34c08980041096068121d : SUCCESS in 10m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e241205c4f784b1fa799a1ee5e973ff5 : SUCCESS in 10m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eb34b773d644005a29fa3ccdc781153 : SUCCESS in 15m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/039c7ab325b3486f8a01f3e0957932c4 : SUCCESS in 1h 07m 17s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/adcdf8c2357d4fec8f17d10d39f1f89e : SUCCESS in 1h 06m 14s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1fd9a0e7e11240988e59f4018f9b61f7 : SUCCESS in 1h 28m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aeb321c8a20548d7b1bc1013f28ef13a : SUCCESS in 59m 12s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0e4084f4a36f482994b157eeee5b0c13 : SUCCESS in 15m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ae692ad5cdf04986a16339cbc2144ebd : SUCCESS in 4m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ea000e29fa1244e5adb8f5e94cb9f7b1 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4f0faeb0b56146de81967a682808102d : SUCCESS in 5m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dab01df9c0854c138999ec96a04132e2 : SUCCESS in 5m 41s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/a025ad13d9594b869a6fab9858c64c9a : SUCCESS in 47m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/12fa20673c794c86a71e51c132714b0d : FAILURE in 5m 47s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9d38ab74cf6244a883aa5046f345896c : SUCCESS in 1h 14m 18s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7e234e41483e42638353f8c13c38c37c : FAILURE in 1h 44m 59s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2d06cfb9f0d44b2e91e0872561a44fdb : SUCCESS in 2h 20m 20s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/46d8e1a810bc47b9bf5e762c52516aa3 : SUCCESS in 1h 06m 57s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/1d6a8ea06a6c4edb95ca067e5fe3d463 : SUCCESS in 1h 37m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/be1c6a48772a406eb03ed056e97359fa : SUCCESS in 41m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c72229889f27475f8ce121653c483969 : SUCCESS in 1h 22m 52s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a13afdfbeae041c1bcc549fc68545953 : SUCCESS in 2h 15m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/feac47d16fa24b7aae68172bd8f7fdf8 : SUCCESS in 49m 29s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/118169dd1e6140fd88bd68f7654051f9 : SUCCESS in 1h 47m 25s","accounts_in_message":[],"_revision_number":1},{"id":"9820b44640a0880977f2d003399afde9697d324f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-17 11:25:34.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/62/873862/1/check/DellEMC_Unity_FC/5f18ce0/DellEMC_Unity_FC/2309 : SUCCESS in 1h 19m 56s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/62/873862/1/check/DellEMC_Unity_iSCSI/5006b68/DellEMC_Unity_iSCSI/2164 : FAILURE in 2h 04m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"7137055711e13f4e4e3a457177e157bb23927528","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-02-17 11:29:29.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_02_2023_11_21_32\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"401d6112baf47a74ab8c857c52aa3a0f49cb2859","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-02-17 11:35:41.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"caef040a8cb1189123f7955b1b6c3acf5a155ddc","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-17 12:23:10.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/873862/1/upstream-check/cinder-tempest-plugin-ontap-fcp/ded7ee5/ : SUCCESS in 57m 39s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/873862/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/8cd7d52/ : SUCCESS in 43m 47s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/873862/1/upstream-check/cinder-tempest-plugin-ontap-nfs/4eaec4e/ : FAILURE in 1h 00m 53s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/873862/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/227f9f2/ : FAILURE in 42m 58s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"ec886ae4508b0ab64c10f101c3667f287134dc3e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-02-17 12:48:42.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/62/873862/1/check/DellEMC_PowerFlex/b50c969/DellEMC_PowerFlex/1626 : SUCCESS in 1h 27m 00s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"2294f0ec8dbedd5003b06ec9afc8241ad14c61ab","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-02-17 12:50:10.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/62/873862/1/check/DellEMC_PowerStore_iSCSI/e5e7b85/DellEMC_PowerStore_iSCSI/1025 : SUCCESS in 1h 15m 31s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/62/873862/1/check/DellEMC_PowerStore_FC/74e2bf6/DellEMC_PowerStore_FC/1160 : SUCCESS in 1h 28m 28s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/62/873862/1/check/DellEMC_PowerStore_NFS/4f76d32/DellEMC_PowerStore_NFS/664 : SUCCESS in 1h 15m 37s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/62/873862/1/check/DellEMC_PowerStore_NVMeTCP/72e4364/DellEMC_PowerStore_NVMeTCP/529 : SUCCESS in 1h 27m 26s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"ab53a1f30ca12cffac45ab38ee7c33f551c7aeb7","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-02-17 13:05:04.000000000","message":"Patch Set 1:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/62/873862/1/fc : FAILURE in 0h 44m 16s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/62/873862/1/iscsi : FAILURE in 1h 04m 33s","accounts_in_message":[],"_revision_number":1},{"id":"71d4a00366383b656140f226d57b8ed0a5d16bf6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-02-17 14:38:15.000000000","message":"Patch Set 1:\n\nSUCCESSFUL build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/1088","accounts_in_message":[],"_revision_number":1},{"id":"32612c6a98894e0073d5fba2ea17755e9ed8cd8a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-17 18:20:37.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-62-873862-1/ns5_nfs : FAILURE in 9h 26m 33s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-62-873862-1/ns5_iscsi : FAILURE in 9h 23m 29s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"7a7305b347cc117ba1a4dbd6b14c8b404dfe56d1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-18 03:09:25.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/45ce37edf3394901b14bb79e141f04d9 : SUCCESS in 1h 59m 54s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"256e2ea696be6b3cf5ca92fc31fb05ec830b432d","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-02-18 23:31:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/873862-1/HPMSA-iSCSI/ : SUCCESS in 1:39:17\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/873862-1/HPMSA-FC/ : SUCCESS in 1:16:56","accounts_in_message":[],"_revision_number":1},{"id":"90739855c570a1574f430688396c6860f0c79de3","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-02-19 17:56:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/873862-1/Seagate-iSCSI/ : SUCCESS in 1:53:25\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/873862-1/Seagate-FC/ : SUCCESS in 1:34:37","accounts_in_message":[],"_revision_number":1},{"id":"69f21285eb2a46182c19a09f1c48cb5756be485e","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-02-19 18:14:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/873862-1/PVME-iSCSI/ : SUCCESS in 1:53:26\n- PVME-FC https://os-logs.seagate.com/dell-ci/873862-1/PVME-FC/ : SUCCESS in 1:34:50","accounts_in_message":[],"_revision_number":1},{"id":"aaf3fd7ccb3616418e5c64b7a9d5e8ce8c5bcc67","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-21 18:47:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5e5b3f3d885b48bc84d057cf16703ad7 : SUCCESS in 1h 12m 23s","accounts_in_message":[],"_revision_number":1},{"id":"db9a9f70bbb7b74eb03d7687c7219addbf41342e","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-02-22 12:28:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"bf72715fc03268f47856152b02ecc90d97a8d678","author":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"date":"2023-02-23 07:09:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eafb7833c7667ad183c910d0b38ad51cec46af7f","author":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"date":"2023-02-23 07:10:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dbe550464cdfa4384cde7d01aa74f169f06b59ed","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-02-23 08:15:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/62/873862/1/20230223121524/NEC-V-ISCSI : SUCCESS in 2h 14m 17s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/62/873862/1/20230223121524/NEC-V-FC : SUCCESS in 4h 59m 45s","accounts_in_message":[],"_revision_number":1},{"id":"3fad974bdaa6726b6412f7abb2b9e7dbd6f00a95","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-02-28 02:40:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/62/873862/1/20230228080835/NEC-ISCSI : SUCCESS in 3h 31m 32s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/62/873862/1/20230228080835/NEC-FC : SUCCESS in 3h 32m 06s","accounts_in_message":[],"_revision_number":1},{"id":"0215cd98596ae1163841c0f13e4de0d5398bb654","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-01 08:10:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"6dbc272d147db4f9c81ed2426dbe600b0dabfcb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"date":"2023-05-16 08:37:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"340cc9b9f664631928be4f30c78e158b8106f521","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"date":"2023-05-16 08:48:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3eb438cc3073a5ce1c253093efc673662f352a2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 09:23:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d2c21ac579143528afcd312ce0e13b2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/caf60413badb41a49529254c2f065a14 : FAILURE in 11m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c884caf404fd47ccb6f81bf064bacc48 : FAILURE in 29m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b32c7e576895427d9ff6376564095e6e : FAILURE in 28m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"16ed30091fe98e354e3bb30fcea20f4aa7141333","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-16 09:31:52.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f12cf8ea857f44e0a802c5c3780cc2d3 : RETRY_LIMIT in 13m 22s","accounts_in_message":[],"_revision_number":3},{"id":"307327c3d9db5df20f7e3f8451bb4a9539f78a2a","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-16 09:45:16.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/d72d982c4dc14f9ca0b5aab007d71526 : POST_FAILURE in 56m 17s","accounts_in_message":[],"_revision_number":3},{"id":"6225e89d30aee4ae7f1f8e84f6f04a2ac0df90c8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-05-16 10:02:39.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/873862 : FAILURE in 1h 10m 16s","accounts_in_message":[],"_revision_number":3},{"id":"a104a55da43cb4a8eb9a0318ab5c870ab4effd25","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-16 10:13:33.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/c18334703a184b299e77eee0fb7153fd\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/124f76875eb14cb99c1533c1b0838c3e : SUCCESS in 1h 22m 33s","accounts_in_message":[],"_revision_number":3},{"id":"e93b4a3b975bf8d882bcaeab97629911629404f8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-16 10:48:10.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/62/873862/3/check/DellEMC_Unity_FC/69b27b5/DellEMC_Unity_FC/146 : SUCCESS in 1h 42m 31s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/62/873862/3/check/DellEMC_Unity_iSCSI/0bf8809/DellEMC_Unity_iSCSI/149 : SUCCESS in 1h 59m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"0de01b953b2c9fadac33516e366bf99fa19632b3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-16 10:53:06.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/62/873862/3/check/iscsi-driver-alletra-6k/55bd611/55bd611.html : FAILURE in 2h 01m 49s","accounts_in_message":[],"_revision_number":3},{"id":"2103cfa66465001299b21c02e3d64f593a205f99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 11:06:31.000000000","message":"Patch Set 3: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffad6db3f6c44cb68e3164c6c81ccd44\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/639c7c44087542fbad13d32a923330ef : FAILURE in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf9a6543f66f4a048efaefbbf22b11b9 : FAILURE in 12m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfdc937fd3434eb99ace72458c7256de : FAILURE in 12m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57ddbbf434914dffa6ade8d2fe601342 : FAILURE in 10m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5567f00273f546259e59ee1599ad3950 : SUCCESS in 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/984ffc3ed92d42bdaadb8eee2f6d78bb : SUCCESS in 1h 10m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/665ca497408d45fda1c4ca98d036f3ea : SUCCESS in 1h 32m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ebb701b40fbc454783530d89d8e96dc4 : SUCCESS in 57m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b83061a19f649b2a68b2adbcd99e83e : SUCCESS in 8m 35s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6ace17df6b3f4599a2d3eeb23beb4f0b : FAILURE in 12m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/335bdfe6772249b18b6c317b7237f128 : SUCCESS in 7m 53s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/517277658345437493af78daf7d0e626 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/852200aafb84436ea18126a7aa9d70f9 : SUCCESS in 5m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f980acf006334a28b187cbfc7255f8d0 : SUCCESS in 8m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e358cf55866d4559b347b1a3227d8ba9 : SUCCESS in 33m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3c3a22f7813b41669ca2056b4f6d7b95 : SUCCESS in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/51db30cf691249f18732f6575760f594 : FAILURE in 1h 51m 45s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/95cd3817de134116866977dced9b9b9f : FAILURE in 1h 33m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c43658cc7d8f418fbc1a79ee9b758ce5 : SUCCESS in 1h 44m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6a59094b79374b9286579b202e08d446 : SUCCESS in 2h 05m 50s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f65ae5f190964a2cb9d35a6727a977e8 : SUCCESS in 1h 01m 14s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/43ef971c0b924c6eb7d946f0782fc294 : SUCCESS in 1h 41m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/75cd93beca94427c9ade69a4ec84154e : SUCCESS in 32m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/479ac3cbcda04a7e92f8cf4813910159 : SUCCESS in 1h 17m 48s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/d2cf28aea1754c92be7dd45b6200a7ed : SUCCESS in 1h 34m 04s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/62194282ba6f4aff955cb3071959586c : SUCCESS in 2h 12m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/903dffc322634e3992114e1a76feaf67 : SUCCESS in 42m 14s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4da089000b6437fba239866c6cc2e61 : SUCCESS in 1h 04m 47s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/057433061598443db5b8695e50d238b1 : FAILURE in 10m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9d596a7e21f0b6068ef572d07b713a027fd5d665","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-05-16 11:34:26.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-62-873862-2/FC/ : SUCCESS in 2h 56m 41s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-62-873862-2/iSCSI/ : SUCCESS in 2h 23m 15s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"34c79fc264bf67cc0c70e4ee3311a2749f2e4367","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-05-16 11:52:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/62/873862/2/20230516173725/NEC-V-ISCSI : SUCCESS in 2h 32m 25s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/62/873862/2/20230516173725/NEC-V-FC : SUCCESS in 3h 14m 36s","accounts_in_message":[],"_revision_number":2},{"id":"6285630e8d7d10a92aa126d2cb71c392e92568ba","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-05-16 12:31:51.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-62-873862-3/FC/ : SUCCESS in 2h 24m 30s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-62-873862-3/iSCSI/ : SUCCESS in 3h 39m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"0c27ec2f297cb5aa98d2281bfe4129de59d2003f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-05-16 12:36:46.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-62-873862-2/ns5_nfs : SUCCESS in 3h 55m 20s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-62-873862-2/ns5_iscsi : SUCCESS in 3h 58m 53s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"cd5c0c06e5d6b5399771727ca686e6b696ef60b6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-05-16 12:53:26.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-62-873862-3/ns5_nfs : SUCCESS in 3h 50m 31s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-62-873862-3/ns5_iscsi : SUCCESS in 3h 53m 4s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"ee1ae949633fc4d3b2938304be80853b0ece4a0b","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-05-16 15:10:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/62/873862/3/20230516204055/NEC-ISCSI : SUCCESS in 3h 12m 45s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/62/873862/3/20230516204055/NEC-FC : SUCCESS in 3h 29m 37s","accounts_in_message":[],"_revision_number":3},{"id":"ee97c3535c76c1874db57d782144915954ed245b","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-05-16 16:59:27.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/62/873862/3/20230516205207/NEC-V-FC : SUCCESS in 2h 57m 42s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/62/873862/3/20230516205207/NEC-V-ISCSI : SUCCESS in 5h 07m 17s","accounts_in_message":[],"_revision_number":3},{"id":"4502733c6505106374d0988815d264fef95ee7be","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-05-17 03:04:33.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/62/873862/3/ : RETRY_LIMIT in 3m 03s","accounts_in_message":[],"_revision_number":3},{"id":"5bf1f68f0a502a19983e32575f77fcfafc6ee63b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"date":"2023-05-17 06:03:24.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ea5d5339f9f7f17c80b29ab23e018e40b362a7aa","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2023-05-17 06:04:42.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"a59747999c45ce3de5214282eda4e3c17dd13eb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"date":"2023-05-17 06:12:41.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"61ea99f7691aaf6df8e0b2915ec534e620d9c89f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 06:40:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e530835b88144749e224d5feb59ec44\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1520dab35fd342dfb2c85d108b90f4b5 : FAILURE in 23m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8d16d6c9f79c4045bc284b0337c1861a : FAILURE in 25m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/498d8c354218465dba800ebe72ca645c : FAILURE in 24m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dd5f2a185126a458fe8dfd4c5653fc211df22414","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-17 06:56:26.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c47d714d74174528acea8facf33abc6b : RETRY_LIMIT in 13m 33s","accounts_in_message":[],"_revision_number":5},{"id":"5d5af34464710d1d1136a978c46d54458c2e67cc","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-05-17 06:58:41.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/873862 : FAILURE in 38m 53s","accounts_in_message":[],"_revision_number":5},{"id":"0e2f94336c79eb33a6806f51b0c1c983b46c090e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-17 07:16:14.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/62/873862/5/check/iscsi-driver-alletra-6k/11e07a4/11e07a4.html : FAILURE in 56m 30s","accounts_in_message":[],"_revision_number":5},{"id":"fe4cf8da1f78be44e5e7729348d093b83fdc2124","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-17 07:28:41.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/e9399b8f6a77463da0375796bbf73900\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/92dc2702fbe344f39daffa8f706f0c61 : FAILURE in 1h 13m 15s","accounts_in_message":[],"_revision_number":5},{"id":"f0ec97ac0907b2ce3ee20d57d5330b50c2d73dc2","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-17 08:17:23.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/b0a0d8035aa048ada3603dc1b7de872c : POST_FAILURE in 1h 19m 39s","accounts_in_message":[],"_revision_number":5},{"id":"ddcd563c56ccce18707e7b41756556f1d2b8e6ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 08:49:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8db04e78b142449c98083208aad069d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79c45460915e4f1f834beb9c39d37075 : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4931123851e42069f880ae24babfe38 : FAILURE in 12m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a236f823a5694744ac713e2d00586b9a : FAILURE in 11m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d49e2e58c0f84ea297251c6e9d386a7e : FAILURE in 12m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/438b6a4914f448e6a0cf2222ce8ed537 : SUCCESS in 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/fe8776febb8d484dbf65295c1b507370 : SUCCESS in 53m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/35f2af88e94b47a3a90868804231b075 : SUCCESS in 1h 09m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5785a12c48340eeaf8b1356c704e136 : SUCCESS in 59m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fd058adabdb44e0957353bb04ece831 : SUCCESS in 8m 24s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/8a95eb35c6a840b5af5d975f3a839bf0 : FAILURE in 14m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b923ab259c7a4701948ab7048ef434d1 : SUCCESS in 4m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/244e2c5a5ebc4a6492926bdbcd0e9ab1 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f91c963fdf9d45dd94c2c34ca21bddc5 : SUCCESS in 6m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fb4a573b2a9f49cc9f598669e3fef5d1 : SUCCESS in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/485d336cd4c64b229a50583aa0f72d10 : SUCCESS in 42m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c07165e47fb74d08b16a685fc4e1ead9 : SUCCESS in 5m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/60b709f8f19940bd84cf14453020c175 : FAILURE in 1h 56m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/85135e10367c4bbabdf92e1723955530 : FAILURE in 1h 38m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/16ad0752c887402daa3d4832de36e4bd : SUCCESS in 1h 31m 09s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/78bbc89450e34e0782edbaab1c95eb30 : SUCCESS in 2h 29m 26s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ab1339c9994841b39b1e8691381c20f0 : SUCCESS in 1h 01m 53s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6a2dbc56e357480aa150c97479ea979e : SUCCESS in 1h 36m 26s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2fd6800bfd6f463ca1969d54744a94bb : SUCCESS in 38m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/80a09cae2a514e40b43cc68ccdaf4bff : SUCCESS in 1h 21m 10s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4d4c697a20b94dc89671499d4422889c : SUCCESS in 1h 30m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a179a11fbb9447bc9a45f5475254b783 : SUCCESS in 1h 53m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ee844c0d2d6469a9190130b6e1cbff0 : SUCCESS in 1h 01m 05s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/477cac84e74b42c9b318ad833b06f0f7 : SUCCESS in 1h 12m 00s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/0909e563dd364d47a2403d82ee87186d : FAILURE in 11m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6aaeebdb575875b9f60fa36cce3fd23f8af539c8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-17 10:03:58.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/62/873862/5/check/DellEMC_Unity_FC/795851f/DellEMC_Unity_FC/149 : FAILURE in 1h 47m 38s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/62/873862/5/check/DellEMC_Unity_iSCSI/85062d4/DellEMC_Unity_iSCSI/152 : SUCCESS in 2h 00m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"7610e77448a0a5c2dcfff8372203dbe888fcb3cf","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-05-17 10:26:34.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-62-873862-4/ns5_nfs : SUCCESS in 4h 12m 35s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-62-873862-4/ns5_iscsi : SUCCESS in 4h 15m 24s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"d8aed67d5200ca8874d9630fa2d8a451b6e370cc","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-05-17 10:54:53.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-62-873862-5/ns5_nfs : SUCCESS in 4h 21m 8s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-62-873862-5/ns5_iscsi : FAILURE in 1h 27m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"9a8c33162dfcf4371c4516a00dbb07ba2625d1e9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-05-17 15:26:06.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/e18ff9b768d94196886b207c2c11acea : RETRY_LIMIT in 1m 00s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/b9911726f3264214a62788a55f62fbad : RETRY_LIMIT in 1m 13s","accounts_in_message":[],"_revision_number":5},{"id":"4574465b4c0252097e40828e55666e74f24edb1c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-05-17 16:12:01.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/62/873862/5/20230517214351/NEC-FC : SUCCESS in 3h 14m 06s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/62/873862/5/20230517214351/NEC-ISCSI : SUCCESS in 3h 28m 07s","accounts_in_message":[],"_revision_number":5},{"id":"04e6f0409f40b0c3e6b86ed59768d6bf14f54df0","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-05-17 18:09:18.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/62/873862/5/20230517211442/NEC-V-ISCSI : SUCCESS in 2h 37m 24s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/62/873862/5/20230517211442/NEC-V-FC : SUCCESS in 5h 54m 33s","accounts_in_message":[],"_revision_number":5},{"id":"53f2c1e31a03a2deb47a21fbc02b2b0e0d8f4105","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-05-18 03:54:27.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/62/873862/5/ : RETRY_LIMIT in 3m 06s","accounts_in_message":[],"_revision_number":5},{"id":"2913b955a8deb73da7cf9ebd87a7545adb42a3f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"date":"2023-05-23 10:15:47.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"3dbbec7b35d0bde900d34572c81fac3de55eeb88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 10:29:08.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dc3034428294c6482e1c31afed4fb1c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/43b0b0808a1a40fe988e61a58fe86f0f : SUCCESS in 12m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a28fef185a4423cbf8927e7413c5eaa : SUCCESS in 10m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/88e6f2146bb84327a66ad498a9cee3be : SUCCESS in 11m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"af49e271065161475b077a7946cc9fab364ecaae","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-23 11:14:52.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7f58a66952b940c9a78fc7a4358f60e4 : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":6},{"id":"64f5a0055728377b15df779eac60b79c28e337d3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-23 11:17:29.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/5a1649b9eb034496a72635451ba51a04\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/35e18393fd7b4256969f06345ccc5238 : FAILURE in 58m 38s","accounts_in_message":[],"_revision_number":6},{"id":"7d887f025b36acee335838ea0b05837c0674e288","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-05-23 11:19:31.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/873862 : FAILURE in 1h 01m 25s","accounts_in_message":[],"_revision_number":6},{"id":"e71d35e8bcd3b8ceca30b92d83dff3b9ffc041fd","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-23 11:42:03.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/62/873862/6/check/iscsi-driver-alletra-6k/be28f47/be28f47.html : FAILURE in 1h 24m 58s","accounts_in_message":[],"_revision_number":6},{"id":"5285607f8e8a91980b3cc2ec29a124106d1c7682","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 12:33:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f1980233ee243b3ac286c978909ced7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a112cfb9f864d7ca24fad56daad79c9 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/639f24f1644f4f77bbba76a01bd0c1fa : SUCCESS in 11m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e29a2af72bbe47f38ec58106a70e6c46 : SUCCESS in 14m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84eb615345bd407183f4197890a171db : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eabe5681a035477b82bbc66844df105c : SUCCESS in 14m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/d842b4d7b95e458091b6a25a1f2e8baf : SUCCESS in 59m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d548f03082584ab8b9baa1bc62ff86da : SUCCESS in 1h 26m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/140eb8b982374c66b9cf24c9f760f1e3 : SUCCESS in 1h 07m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f2ac652d1274fa5b3bfd307ff693942 : SUCCESS in 8m 05s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9289015ed0f648f1ab88ef6f9a11e81f : SUCCESS in 15m 56s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bfc7fd9f2c0e4fc4bf46f5affd1cc3f6 : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3f012a6c0c984b309e9049293a266e37 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3ae7b945a1a04030ad7780b75c616c75 : SUCCESS in 6m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9f986491610d46b894710f066e25486c : SUCCESS in 5m 48s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b1c2951567a340539c509c32bc8c83e2 : SUCCESS in 36m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/616a5d8e5c844c78bd26828070a4006c : SUCCESS in 6m 12s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/67a7d28569c04778816da6bc748d062f : SUCCESS in 1h 35m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/65b81f3ee851431383bd373e2c669b82 : FAILURE in 1h 56m 33s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c410f4958cb749169a3f4e1eb9331b6f : SUCCESS in 1h 34m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/eb5fc089a975467cb8d0c18107c94fe4 : POST_FAILURE in 8m 33s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/3876aeb6033a449ba9b2839a780b84e4 : SUCCESS in 1h 11m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a87af302dca8478889c579fc65b3a18d : SUCCESS in 1h 20m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/13232b7c61f24da0b31eb4ba2eda4a3f : SUCCESS in 35m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b5531b36aff84d6f80e26f5f67630a2b : SUCCESS in 1h 06m 09s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/19dc057db73f4246b05311f9599cbb75 : FAILURE in 7m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b1140c671a2f4630aa5fc98873e9b43c : SUCCESS in 2h 14m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ff23c0b79504fa887130fe299daf867 : SUCCESS in 44m 59s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/363d5c7995f846b1a1376c2530b6af09 : SUCCESS in 1h 20m 06s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/5b25f2ff9988406d9433cf7d411d196a : SUCCESS in 9m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"46e519130832ae33df3f93e633eebdd6c7c0a6ca","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-05-23 12:34:40.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"fd406f93128ab80a5920ce2fda62d01c3584285e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-05-23 12:41:34.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_05_2023_12_34_46\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"63bdf2f1cccb924ccff29c2c1a8dcb97ec7e6f2e","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-05-23 13:00:59.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"6f392486568a3791ed5a96e5833596ae4b4cab76","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-05-23 13:03:45.000000000","message":"Patch Set 6:\n\nBuild failed.","accounts_in_message":[],"_revision_number":6},{"id":"59ec989f4225a9dd595eccc62104b72dbda75061","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-23 13:14:19.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/d74d7d4092ef419dbcf77766e58a4a18 : POST_FAILURE in 55m 51s","accounts_in_message":[],"_revision_number":6},{"id":"d71e8ed77b3511e6a0f39c2cb268b6d184b1efcf","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-23 13:20:31.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/873862/6/upstream-check/cinder-tempest-plugin-ontap-fcp/a2072e6/ : RETRY_LIMIT in 8m 07s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/873862/6/upstream-check/cinder-tempest-plugin-ontap-iscsi/b19afc7/ : SUCCESS in 40m 47s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/873862/6/upstream-check/cinder-tempest-plugin-ontap-nfs/4aec118/ : SUCCESS in 41m 54s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/62/873862/6/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/95b0a1d/ : SUCCESS in 44m 25s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"3f35d8b32e99b23b8f2c1341383943eb9b1b0c85","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-05-23 14:02:54.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-62-873862-6/ns5_nfs : SUCCESS in 3h 42m 48s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-62-873862-6/ns5_iscsi : SUCCESS in 3h 46m 37s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"bf6735925e980cbc211d26a3ad2464308e0f0576","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-05-23 14:11:52.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/62/873862/6/check/DellEMC_PowerFlex/3da60a9/DellEMC_PowerFlex/99 : SUCCESS in 1h 35m 16s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"2129cf0f85e63c88468ef65cd2c686ffb1a56f87","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-05-23 16:26:17.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-62-873862-6/FC/ : SUCCESS in 2h 28m 17s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-62-873862-6/iSCSI/ : FAILURE in 6h 10m 7s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"6f9f49b80f4bbea1e73149e0a902f1f4b0a3cdfc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-23 17:36:15.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/62/873862/6/check/DellEMC_Unity_FC/a61e069/DellEMC_Unity_FC/170 : FAILURE in 1h 49m 18s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/62/873862/6/check/DellEMC_Unity_iSCSI/f6868cd/DellEMC_Unity_iSCSI/170 : FAILURE in 2h 04m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"61398103f7e3739fb0f1b9e573667fff2eaf2b85","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-05-23 17:43:06.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/62/873862/6/check/DellEMC_PowerStore_iSCSI/6409b87/DellEMC_PowerStore_iSCSI/55 : SUCCESS in 51m 10s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/62/873862/6/check/DellEMC_PowerStore_FC/d04deb5/DellEMC_PowerStore_FC/55 : SUCCESS in 1h 15m 11s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/62/873862/6/check/DellEMC_PowerStore_NFS/bdbb3ad/DellEMC_PowerStore_NFS/54 : SUCCESS in 1h 05m 45s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/62/873862/6/check/DellEMC_PowerStore_NVMeTCP/727b489/DellEMC_PowerStore_NVMeTCP/54 : SUCCESS in 1h 05m 37s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"a696e4740e3d4e5ea76b40925ff1bbb95975a5fe","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-05-23 20:48:44.000000000","message":"Patch Set 6:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1367/","accounts_in_message":[],"_revision_number":6},{"id":"2360a881d07afd29719b4fe0e29b49115a010016","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-05-24 00:45:19.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/62/873862/6/20230524061805/NEC-V-ISCSI : SUCCESS in 2h 53m 48s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/62/873862/6/20230524061805/NEC-V-FC : SUCCESS in 3h 27m 11s","accounts_in_message":[],"_revision_number":6},{"id":"03bb5bf234c24fd1991bebe926fecf45853e7812","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-05-24 02:47:26.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/62/873862/6/check/DellEMC_PowerMAX_FC/b42df19/DellEMC_PowerMAX_FC/76 : SUCCESS in 6h 07m 25s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/62/873862/6/check/DellEMC_PowerMAX_iSCSI/5dadb36/DellEMC_PowerMAX_iSCSI/74 : SUCCESS in 5h 36m 24s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"8887dcc272a761c69e309708a6209fb5f9e1aa4d","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-05-24 05:54:03.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"066b4bf1428c96d1eab3025d271ab56a1341f2f4","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-05-24 08:06:14.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/62/873862/6/ : RETRY_LIMIT in 11m 20s","accounts_in_message":[],"_revision_number":6},{"id":"ccf338a75401a3f38ca949fdca6dd06e6da45e78","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-05-24 10:58:02.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/62/873862/6/20230524160448/NEC-ISCSI : SUCCESS in 3h 41m 37s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/62/873862/6/20230524160448/NEC-FC : SUCCESS in 3h 53m 10s","accounts_in_message":[],"_revision_number":6},{"id":"6aaa067617390699c02110f6fa850fa94d4d0433","author":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"date":"2023-07-17 05:52:13.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"053d243ff3ad4d87679b5161a55a75c8d3cbc2e6","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-18 04:00:09.000000000","message":"Patch Set 6:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/62/873862/6/fc : SUCCESS in 0h 46m 13s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/62/873862/6/iscsi : SUCCESS in 0h 58m 17s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"cfeed78749c9b88e97f5a29e0750235a52266dd1","revisions":{"dcafb8c28160a5883a04342d5a2b12cd026b6255":{"kind":"REWORK","_number":1,"created":"2023-02-17 08:53:33.000000000","uploader":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"ref":"refs/changes/62/873862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/873862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/873862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/873862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/873862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/873862/1"}}},"commit":{"parents":[{"commit":"d6fd72b47d2b19794afb0865f4c9cdabc299788b","subject":"Merge \"[SVf] As part of Flashcopy 2.0 adding config parameter to support volumegroup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6fd72b47d2b19794afb0865f4c9cdabc299788b"}]}],"author":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2023-02-17 08:33:17.000000000","tz":480},"committer":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2023-02-17 08:52:50.000000000","tz":480},"subject":"cinder-backup:use snapshot_id create backup","message":"cinder-backup:use snapshot_id create backup\n\nAfter the backup service is down,use the body contain snapshot_id to create a backup,\nThe cinder_scheduler will detect that the backup service is down, and enter the exception processing phase. At this time, update the volume status to the previous status. Because the previous status of the volume is not assigned before the exception, the default is None, so the volume status is changed to None.\n\nThis patch ensure that the volume state is not changed when creating a backup from a snapshot.\nCloses-bug: #2007631\n\nChange-Id: I99edfa2b18f3315d7ff81981216dccb8e32e0fe4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dcafb8c28160a5883a04342d5a2b12cd026b6255"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dcafb8c28160a5883a04342d5a2b12cd026b6255"}]},"branch":"refs/heads/master"},"08b918cf955864232588ef5471e610950ce59cab":{"kind":"REWORK","_number":2,"created":"2023-05-16 08:37:22.000000000","uploader":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"ref":"refs/changes/62/873862/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/873862/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/873862/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/873862/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/873862/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/873862/2"}}},"commit":{"parents":[{"commit":"d6fd72b47d2b19794afb0865f4c9cdabc299788b","subject":"Merge \"[SVf] As part of Flashcopy 2.0 adding config parameter to support volumegroup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6fd72b47d2b19794afb0865f4c9cdabc299788b"}]}],"author":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2023-02-17 08:33:17.000000000","tz":480},"committer":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2023-05-16 08:36:46.000000000","tz":480},"subject":"cinder-backup:use snapshot_id create backup","message":"cinder-backup:use snapshot_id create backup\n\nAfter the backup service is down,use the body contain snapshot_id to create a backup,\nThe cinder_scheduler will detect that the backup service is down, and enter the exception processing phase. At this time, update the volume status to the previous status. Because the previous status of the volume is not assigned before the exception, the default is None, so the volume status is changed to None.\n\nThis patch ensure that the volume state is not changed when creating a backup from a snapshot.\nCloses-bug: #2007631\n\nChange-Id: I99edfa2b18f3315d7ff81981216dccb8e32e0fe4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08b918cf955864232588ef5471e610950ce59cab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/08b918cf955864232588ef5471e610950ce59cab"}]},"branch":"refs/heads/master"},"19076c6dc2545b7a4c7b3cf9213c616f4c6f050f":{"kind":"REWORK","_number":3,"created":"2023-05-16 08:48:12.000000000","uploader":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"ref":"refs/changes/62/873862/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/873862/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/873862/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/873862/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/873862/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/873862/3"}}},"commit":{"parents":[{"commit":"d6fd72b47d2b19794afb0865f4c9cdabc299788b","subject":"Merge \"[SVf] As part of Flashcopy 2.0 adding config parameter to support volumegroup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6fd72b47d2b19794afb0865f4c9cdabc299788b"}]}],"author":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2023-02-17 08:33:17.000000000","tz":480},"committer":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2023-05-16 08:47:48.000000000","tz":480},"subject":"cinder-backup:use snapshot_id create backup","message":"cinder-backup:use snapshot_id create backup\n\nAfter the backup service is down,use the body contain snapshot_id to create a backup,\nThe cinder_scheduler will detect that the backup service is down, and enter the exception processing phase. At this time, update the volume status to the previous status. Because the previous status of the volume is not assigned before the exception, the default is None, so the volume status is changed to None.\n\nThis patch ensure that the volume state is not changed when creating a backup from a snapshot.\nCloses-bug: #2007631\n\nChange-Id: I99edfa2b18f3315d7ff81981216dccb8e32e0fe4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/19076c6dc2545b7a4c7b3cf9213c616f4c6f050f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/19076c6dc2545b7a4c7b3cf9213c616f4c6f050f"}]},"branch":"refs/heads/master"},"7021b6adc7dccd5ab6b47b8fd5af11cc96783150":{"kind":"REWORK","_number":4,"created":"2023-05-17 06:03:24.000000000","uploader":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"ref":"refs/changes/62/873862/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/873862/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/873862/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/873862/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/873862/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/873862/4"}}},"commit":{"parents":[{"commit":"d6fd72b47d2b19794afb0865f4c9cdabc299788b","subject":"Merge \"[SVf] As part of Flashcopy 2.0 adding config parameter to support volumegroup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6fd72b47d2b19794afb0865f4c9cdabc299788b"}]}],"author":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2023-02-17 08:33:17.000000000","tz":480},"committer":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2023-05-17 06:02:58.000000000","tz":480},"subject":"cinder-backup:use snapshot_id create backup","message":"cinder-backup:use snapshot_id create backup\n\nAfter the backup service is down,use the body contain snapshot_id to create a backup,\nThe cinder_scheduler will detect that the backup service is down, and enter the exception processing phase. At this time, update the volume status to the previous status. Because the previous status of the volume is not assigned before the exception, the default is None, so the volume status is changed to None.\n\nThis patch ensure that the volume state is not changed when creating a backup from a snapshot.\nCloses-bug: #2007631\n\nChange-Id: I99edfa2b18f3315d7ff81981216dccb8e32e0fe4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7021b6adc7dccd5ab6b47b8fd5af11cc96783150"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7021b6adc7dccd5ab6b47b8fd5af11cc96783150"}]},"branch":"refs/heads/master"},"3ad12498b2ea02033a2282a92c791879499ed7bb":{"kind":"REWORK","_number":5,"created":"2023-05-17 06:12:41.000000000","uploader":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"ref":"refs/changes/62/873862/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/873862/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/873862/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/873862/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/873862/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/873862/5"}}},"commit":{"parents":[{"commit":"d6fd72b47d2b19794afb0865f4c9cdabc299788b","subject":"Merge \"[SVf] As part of Flashcopy 2.0 adding config parameter to support volumegroup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6fd72b47d2b19794afb0865f4c9cdabc299788b"}]}],"author":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2023-02-17 08:33:17.000000000","tz":480},"committer":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2023-05-17 06:12:17.000000000","tz":480},"subject":"cinder-backup:use snapshot_id create backup","message":"cinder-backup:use snapshot_id create backup\n\nAfter the backup service is down,use the body contain snapshot_id to create a backup,\nThe cinder_scheduler will detect that the backup service is down, and enter the exception processing phase. At this time, update the volume status to the previous status. Because the previous status of the volume is not assigned before the exception, the default is None, so the volume status is changed to None.\n\nThis patch ensure that the volume state is not changed when creating a backup from a snapshot.\nCloses-bug: #2007631\n\nChange-Id: I99edfa2b18f3315d7ff81981216dccb8e32e0fe4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3ad12498b2ea02033a2282a92c791879499ed7bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3ad12498b2ea02033a2282a92c791879499ed7bb"}]},"branch":"refs/heads/master"},"cfeed78749c9b88e97f5a29e0750235a52266dd1":{"kind":"REWORK","_number":6,"created":"2023-05-23 10:15:47.000000000","uploader":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"ref":"refs/changes/62/873862/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/62/873862/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/873862/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/873862/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/62/873862/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/62/873862/6"}}},"commit":{"parents":[{"commit":"d6fd72b47d2b19794afb0865f4c9cdabc299788b","subject":"Merge \"[SVf] As part of Flashcopy 2.0 adding config parameter to support volumegroup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d6fd72b47d2b19794afb0865f4c9cdabc299788b"}]}],"author":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2023-02-17 08:33:17.000000000","tz":480},"committer":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2023-05-23 10:15:19.000000000","tz":480},"subject":"cinder-backup:use snapshot_id create backup","message":"cinder-backup:use snapshot_id create backup\n\nAfter the backup service is down,use the body contain snapshot_id to create a backup,\nThe cinder_scheduler will detect that the backup service is down, and enter the exception processing phase. At this time, update the volume status to the previous status. Because the previous status of the volume is not assigned before the exception, the default is None, so the volume status is changed to None.\n\nThis patch ensure that the volume state is not changed when creating a backup from a snapshot.\nCloses-bug: #2007631\n\nChange-Id: I99edfa2b18f3315d7ff81981216dccb8e32e0fe4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cfeed78749c9b88e97f5a29e0750235a52266dd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cfeed78749c9b88e97f5a29e0750235a52266dd1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
