)]}'
{"id":"openstack%2Fcinder~874242","triplet_id":"openstack%2Fcinder~master~Ia1b92341fc44fa903c238b2cb82496b77012a367","project":"openstack/cinder","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-07-03 19:31:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia1b92341fc44fa903c238b2cb82496b77012a367","subject":"db: Remove the legacy \u0027migration_version\u0027 table","status":"MERGED","created":"2023-02-17 12:57:16.000000000","updated":"2023-07-03 20:00:31.000000000","submitted":"2023-07-03 19:31:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874242-sqlalchemy-20","meta_rev_id":"42f43c7f6c5dc649aad9d348a207c07686a329b4","_number":874242,"virtual_id_number":874242,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-03 19:31:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-06-30 18:34:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-06-26 07:55:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":2,"date":"2023-05-23 09:43:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2023-06-30 18:34:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}],"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-17 12:57:18.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-02-17 13:06:15.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-02-17 13:19:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-17 14:08:18.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-17 14:10:16.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-17 15:18:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-17 17:25:32.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-17 17:53:18.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-02-17 18:31:05.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-02-17 19:30:30.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-17 19:40:27.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-02-17 21:45:25.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-02-17 22:51:46.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-02-17 22:58:11.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-17 23:05:03.000000000","updated_by":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"reviewer":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"state":"CC"},{"updated":"2023-02-18 02:43:04.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-19 14:52:37.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-02-20 15:12:29.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-02-20 15:14:34.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2023-02-21 20:07:33.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-23 18:30:40.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-02-27 05:12:23.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"state":"REVIEWER"},{"updated":"2023-02-28 20:17:11.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-17 05:38:21.000000000","updated_by":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"reviewer":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"state":"CC"},{"updated":"2023-03-23 16:47:00.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-03-23 16:47:19.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2023-05-23 09:43:48.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-06-26 07:55:02.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-06-26 12:30:30.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-06-27 01:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-06-30 18:34:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2023-07-03 12:54:41.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"}],"messages":[{"id":"6fbf8a18ab8131ca4b361822b81036f249e51208","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-17 12:57:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37b30ef34211a22e61f9ed0a8a18f145c603153a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-02-17 12:57:18.000000000","message":"Patch Set 1:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":1},{"id":"dd269111443abd57dfa64312407bfda2ae488636","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-17 13:06:15.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"41ad3b203747a185ea53d87172c3ded3b6ae5c84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 13:19:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c938f86bd80b43f79db4433e0d7b06fe\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0442c6402aab4d8eb4aaac1e6a74020a : SUCCESS in 20m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1a0443197bd545ef8479df7e5ab5361b : SUCCESS in 8m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e9d705f31b27c073c01c5e68b81b5e363155227","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-17 14:08:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/874242 : FAILURE in 23m 27s","accounts_in_message":[],"_revision_number":1},{"id":"468c119abf5ac564748360f489988f529da4ad58","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-17 14:10:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/42/874242/1/ : FAILURE in 1h 04m 50s","accounts_in_message":[],"_revision_number":1},{"id":"f442ec93be2f07c75d3b60a30ae60c3ecd22c862","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-17 14:10:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/42/874242/1/check/iscsi-driver-alletra-6k/cbea170/cbea170.html : FAILURE in 25m 59s","accounts_in_message":[],"_revision_number":1},{"id":"2ffec987284bcb7efa0765739fdab9fe4cb8762a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-17 14:20:10.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/874242 : FAILURE in 26m 09s","accounts_in_message":[],"_revision_number":1},{"id":"017fcb3f778860ebf2026b8da3c60f5a68a55c01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 15:18:45.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/13b016dc02d0456fb0ee1d8697ba9be3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2bb318ac5c54999a6d2da3a5e27f017 : FAILURE in 9m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4127cf0ca114334b63abadee7151598 : SUCCESS in 11m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7360b8865af04dfab593118ff0829d60 : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42b48b4c214143b2a4f047951501ef84 : SUCCESS in 12m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/ae96ae9493144ea8b821fa7f303ae029 : SUCCESS in 59m 37s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ccd23424f7c94b1a8f4ebcd8e626e794 : SUCCESS in 54m 46s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c8e32c5683e5439c9bb57456b366e65e : SUCCESS in 1h 16m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d415a0f5d8d2425a84f18cdb036347b7 : SUCCESS in 46m 45s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/624594af84f4477f9e6ca20aeb29c7aa : SUCCESS in 19m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4d57b374260842c6a1ac4585110deffa : SUCCESS in 5m 14s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1ed45b24330f400dbf4d83d2723333cb : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f7b6c556979c4e6b9803311f37a32fa0 : SUCCESS in 5m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6387819e32fb45839d0ccd528ae47c81 : SUCCESS in 6m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0c34b7a8236748b39880fc94e0c41c81 : SUCCESS in 40m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b72c7cd9c32d4312a4831cb6a6ee3b55 : FAILURE in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/02b7cf04a1944b33814d757eb29d0576 : SUCCESS in 1h 20m 20s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bfefa4779bd84ae88f827b428fe579ee : FAILURE in 1h 38m 46s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f762e995e7ae4b5ca68e362dbf15dcd7 : SUCCESS in 1h 36m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/ab109f5c61f04d4e843256172bd96192 : SUCCESS in 1h 08m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/2fcc24b388714f7390e550699d699729 : SUCCESS in 1h 42m 43s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9ef5e115292c4568a66333ac3bca9935 : SUCCESS in 42m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/ad9214e842e74466818ed280a2ec453e : SUCCESS in 1h 22m 47s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/006d2570b0274c11bf144c1ac54f81bf : SUCCESS in 2h 14m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d15528f055d4521a57af7d9192e745e : SUCCESS in 1h 06m 05s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66cda7978eb34e998d569342917755ba : SUCCESS in 1h 29m 12s","accounts_in_message":[],"_revision_number":1},{"id":"30cf5968c1bfc8c781011c3f08cd8d64d93ecb08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-17 15:44:28.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"43072c3621a9d994a955277e44a3ccd3abf2c782","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-02-17 15:44:31.000000000","message":"Patch Set 2:\n\nNo Builds Executed \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2},{"id":"756772ff6deed34a680b46dbdfef209310ff6316","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 15:54:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5d7d4f230304fea878eccf22073d6cc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/697402b34ecb43c1abd99840f568acd3 : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/09799a6ca98241c68af5cc446d623031 : SUCCESS in 8m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9de26845c209ca5f23ee5973c17eb1da8382b26e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-17 16:22:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/874242 : FAILURE in 23m 28s","accounts_in_message":[],"_revision_number":2},{"id":"1df9de5ef15c8eb5d8813df23ea31da9361e2984","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-17 16:23:57.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/42/874242/2/check/iscsi-driver-alletra-6k/37799b4/37799b4.html : FAILURE in 26m 07s","accounts_in_message":[],"_revision_number":2},{"id":"8fa10e4434fa9cabb172c49de5300fc0aff242c7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-17 16:25:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/874242 : FAILURE in 25m 59s","accounts_in_message":[],"_revision_number":2},{"id":"4c30b3383a107616841e532b3b319b6d0b9c2d25","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-17 16:50:36.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/42/874242/2/ : FAILURE in 57m 41s","accounts_in_message":[],"_revision_number":2},{"id":"844aa7d3bdcfd63fb6e434297e7c024772540ad2","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-02-17 17:25:32.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/42/874242/2/check/fujitsu-eternus-dx-iscsi/2196723 : SUCCESS in 1h 38m 22s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/42/874242/2/check/fujitsu-eternus-dx-fc/2810a0e : FAILURE in 3m 14s","accounts_in_message":[],"_revision_number":2},{"id":"7ba71b70296b527355296a2a50dafd7209c7600f","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-17 17:53:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/7ee99f4741e94ba2bd07bbc61ce36b0c : SUCCESS in 1h 03m 50s","accounts_in_message":[],"_revision_number":2},{"id":"b61ab8aefce87670d78665d38323dceaaed70691","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 18:22:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e72907610c645a88251e319b6b05f8f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12b74aeaa4424a53b4622ebc798a227e : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2df779fff7f44cbbd0aa88fe38a268a : SUCCESS in 9m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39e26623e86d49b585b30731d9482ee7 : SUCCESS in 11m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b5120903b144b37821c7f479130f1ed : SUCCESS in 12m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/0a263dd2936f4ca89790e50440fa5866 : SUCCESS in 58m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0e3c17e2f4c44feba2d23cb15742dac8 : SUCCESS in 46m 07s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/52c148bd28c6474b94996654dfe943e7 : SUCCESS in 1h 22m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e196d296598d4dbba7bae7da7b30ba5a : SUCCESS in 53m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0c74f2d4405d4958ab4cd741f2e4d028 : SUCCESS in 17m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/91aa0626e9294e35adae3cdcefc04688 : SUCCESS in 5m 50s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fd7469f3b94246f1b6c7206840542ed1 : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a19ff71dfa1f4f5e9258ce34cdbe4a57 : SUCCESS in 6m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7b5afe0f6fb54f21a52d6e4852c610ce : SUCCESS in 5m 55s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/878bcba1b9ca42e08db7bf3642d54b8c : SUCCESS in 53m 40s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5cece28610ec4fd7aa1304068a567d3f : FAILURE in 5m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4bc064d0df534b60b12469accd8af705 : SUCCESS in 1h 44m 56s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/c837b4cbafd8450b9ee6c7320343e530 : FAILURE in 1h 46m 39s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c9b85820753643f5b6799b3a33760df3 : SUCCESS in 1h 50m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d01a9e6bcc0a466380772c4e56cbe2f2 : SUCCESS in 1h 16m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/781b18290d754ab9b3fb19265c5ee50d : SUCCESS in 1h 15m 47s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bcfb05b0efcd4d60a81f0aec7066c6ed : SUCCESS in 1h 02m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0ac2dbf9e81c49a981d409bb76d8848c : SUCCESS in 1h 48m 09s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/190a71fc5cdb414eb12a4b314101bbe2 : SUCCESS in 2h 19m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ba08e849a2b4a7fbbc16ded05a85fbf : SUCCESS in 1h 02m 30s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26add69cd7cf435d85728a25e2ac2004 : SUCCESS in 1h 22m 48s","accounts_in_message":[],"_revision_number":2},{"id":"aab70955dfad27ca063e74b86729daa912dc0977","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-02-17 18:31:05.000000000","message":"Patch Set 2:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/17_02_2023_18_23_04\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"204aeb975e87fc34341dd40928fe62a9f2a0f7ea","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-17 19:30:30.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"1a30a87d2723bcbd3a361e2061459ecb4357117e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-17 19:40:27.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/42/874242/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/27c1e96/ : SUCCESS in 1h 48m 52s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/42/874242/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/a4fbc29/ : FAILURE in 2h 27m 41s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/42/874242/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/c8b0f74/ : SUCCESS in 1h 50m 02s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/42/874242/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/6c473f1/ : FAILURE in 2h 24m 03s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy pure-devstack-cinder-tempest-nvme-roce-aio-jammy : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0a12bfaa2bf37833cf74f990b078d3b21d03c033","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-17 21:45:25.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/42/874242/2/check/DellEMC_Unity_FC/3fa235d/DellEMC_Unity_FC/2316 : SUCCESS in 1h 43m 57s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/42/874242/2/check/DellEMC_Unity_iSCSI/28a5c7b/DellEMC_Unity_iSCSI/2170 : SUCCESS in 1h 36m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"4475498e8332258d74c921e08645a400b246c5f8","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-02-17 22:51:46.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/42/874242/2/check/DellEMC_PowerStore_iSCSI/8436e71/DellEMC_PowerStore_iSCSI/1030 : SUCCESS in 1h 14m 24s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/42/874242/2/check/DellEMC_PowerStore_FC/d8cc56e/DellEMC_PowerStore_FC/1165 : SUCCESS in 1h 28m 35s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/42/874242/2/check/DellEMC_PowerStore_NFS/fa45caf/DellEMC_PowerStore_NFS/669 : SUCCESS in 1h 19m 03s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/42/874242/2/check/DellEMC_PowerStore_NVMeTCP/42b1923/DellEMC_PowerStore_NVMeTCP/534 : SUCCESS in 1h 25m 50s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"83df1312d89fe4280c1cb077284e2959cc56c2e5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-02-17 22:58:11.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/42/874242/2/check/DellEMC_PowerFlex/e38c053/DellEMC_PowerFlex/1631 : SUCCESS in 1h 18m 36s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"eaa097e4da3ba1c0e675ff0f374b13a61607ffb0","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-02-17 23:05:03.000000000","message":"Patch Set 2:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/42/874242/2/fc : FAILURE in 0h 53m 13s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/42/874242/2/iscsi : FAILURE in 1h 04m 13s","accounts_in_message":[],"_revision_number":2},{"id":"833616a90689d267b70ef71a582cdc18022df752","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-18 02:43:04.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-42-874242-1/ns5_nfs : FAILURE in 1h 1m 9s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-42-874242-1/ns5_iscsi : FAILURE in 12h 17m 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"3ffb98a14a466ba121b98d185971660eb97388f9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-18 04:24:06.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-42-874242-2/ns5_nfs : FAILURE in 11h 58m 21s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-42-874242-2/ns5_iscsi : FAILURE in 12h 16m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"f2f6fbd570a8e1b97978da0a7528fc3c5a1f1243","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-02-19 14:52:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/874242-2/HPMSA-iSCSI/ : SUCCESS in 1:38:18\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/874242-2/HPMSA-FC/ : FAILURE in 1:18:06","accounts_in_message":[],"_revision_number":2},{"id":"78978ea272a574308004dd595b6152cca1c280b2","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-02-20 15:12:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/874242-2/PVME-iSCSI/ : FAILURE in 1:51:19\n- PVME-FC https://os-logs.seagate.com/dell-ci/874242-2/PVME-FC/ : SUCCESS in 1:28:10","accounts_in_message":[],"_revision_number":2},{"id":"684cd57d90c64a9e5545d99dfd587bc3008b83e0","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-02-20 15:14:34.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/874242-2/Seagate-iSCSI/ : SUCCESS in 1:53:42\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/874242-2/Seagate-FC/ : SUCCESS in 1:31:58","accounts_in_message":[],"_revision_number":2},{"id":"1ff9fc60f8f36b08ce2b07491c32a8d22fec66e1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-21 20:07:33.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/abc3e33f70aa4b7fbd07ff5a2fc48218 : SUCCESS in 1h 10m 49s","accounts_in_message":[],"_revision_number":2},{"id":"a238d0253bf374c4c85abd051d8ff373cc9b2dbe","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-02-23 18:30:40.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/42/874242/2/20230224010747/NEC-V-FC : SUCCESS in 2h 09m 46s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/42/874242/2/20230224010747/NEC-V-ISCSI : SUCCESS in 2h 22m 50s","accounts_in_message":[],"_revision_number":2},{"id":"cd208139ecef47f88db6efd4ea5b7cfbb933e5ef","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2023-02-27 05:12:23.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"750166d7e1e19f8a11d390a560f25532bafa7425","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-02-28 20:17:11.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/42/874242/2/20230228221459/NEC-FC : SUCCESS in 3h 31m 13s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/42/874242/2/20230228221459/NEC-ISCSI : SUCCESS in 7h 02m 08s","accounts_in_message":[],"_revision_number":2},{"id":"de9ca5376e7bc5192b5d09f36a165314935938e2","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-01 08:03:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"b0d47234cff89c9732c577a0d544c8362aaaaff9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-03-16 09:18:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f9c0cd69a6abe1a1d1d66cc2c0b5dd33525b9fe5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-16 09:32:27.000000000","message":"Patch Set 3:\n\nBuild Started http://openstack-ci.spb.yadro.com:8080/job/openstack-gerrit-trigger-fc/126/ (2/2)","accounts_in_message":[],"_revision_number":3},{"id":"e557ea4c73a6e60df4c7f0758d980167d9c558fa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-16 09:39:33.000000000","message":"Patch Set 3:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/1189","accounts_in_message":[],"_revision_number":3},{"id":"ee72cca7a002c3800394198e32fd4206df43a002","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-16 09:47:52.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder finger://openstack-ci-fc/4f2a89d7b5e84765b45770aa8d4adee3 : RETRY_LIMIT in 2m 15s","accounts_in_message":[],"_revision_number":3},{"id":"559b016a53b940193ff157771690a40dd1374571","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 09:54:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d6289a762b54437984da88443fcb15f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ac58ec0dc5a4db28d151907c008c4b4 : FAILURE in 33m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f1b51fbee8e14a1684c329d3de160327 : FAILURE in 29m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"673f4e8a3e96e0023020cb538c3ebd1d6eca3a47","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-16 09:58:19.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c5f8a205e2c04bfeb3726862400baa7e : FAILURE in 36m 56s","accounts_in_message":[],"_revision_number":3},{"id":"2d4d2390875ba155daefe21858af3ae78a8c5861","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 10:21:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4084e3e0bd740308317abe6b6321450\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3805f58922940f9b7e906923771a68f : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/660ad656998b47639ce93a59e2685729 : FAILURE in 17m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78209b35dadc4051a3241e3847ceb186 : FAILURE in 17m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f17545b77d9044a2b4834be4eadc0058 : SUCCESS in 13m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/20f1613144ec41e5984261fe13c0e8d3 : FAILURE in 49m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fb337d33843a4311abc2550901e9976b : FAILURE in 49m 00s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/75dfabbf91f04ac8ad804998aae9e1e8 : FAILURE in 16m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba6ddafb3f274775af40a1eb87a4d571 : POST_FAILURE in 28m 32s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f28f013d91e843ce9ea706d541f66b0a : FAILURE in 17m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d325c49e35fd4c9aa9d072429c316a11 : SUCCESS in 5m 09s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fd62246d00b64494a26059489b3c3aaf : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/edf90a50d7434745ad5e46321cb64037 : FAILURE in 5m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/68b483b7998a43d2a61153eac5df6c93 : FAILURE in 5m 04s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/915a4e778c95419ba369facfbbe76b7b : FAILURE in 28m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3ab4362bb43f49dbb0f32af76361fdd6 : FAILURE in 6m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/283e11653c2747aebc812accfbbb722e : FAILURE in 30m 46s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/71eb7a2f5a5549e18e29000d9f32020c : FAILURE in 33m 08s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/c65f57ebb07a46a7a2fd48ce02eea5d4 : FAILURE in 25m 12s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0c7c915eba2b4df4a864e47092576c5c : FAILURE in 50m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ad856f7cd8f045acb25e3ea2815fc1da : FAILURE in 28m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/faec953959cc4ad9b27aa2e4d00f32df : FAILURE in 15m 04s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/14df3bbb85ad4658a3bc5851870e5946 : FAILURE in 34m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/75e4bb2bed2b4a8a9c50abaa1695c9b3 : FAILURE in 21m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2de9d9569a64cb28fbbb4973d639d71 : FAILURE in 19m 38s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08f99ed3071d48c1831c11127f79eaac : FAILURE in 20m 55s","accounts_in_message":[],"_revision_number":3},{"id":"1a450ea8d919f728d3e2490f4870a1de8faa1376","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-16 10:21:32.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/874242 : FAILURE in 27m 02s","accounts_in_message":[],"_revision_number":3},{"id":"a23756bfa82517daa3b18f4ec598d098a54529b8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-16 10:32:14.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-42-874242-3/ns5_nfs : FAILURE in 28m 0s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-42-874242-3/ns5_iscsi : FAILURE in 27m 50s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"0f948f3a8ec9dc4c4f35ed0945a9c45482ffcf09","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-16 12:01:43.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/42/874242/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/a2509ed/ : FAILURE in 33m 16s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/42/874242/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6b33acc/ : FAILURE in 31m 21s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/42/874242/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/40850e2/ : FAILURE in 32m 17s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/42/874242/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/2584c57/ : FAILURE in 29m 42s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy pure-devstack-cinder-tempest-nvme-roce-aio-jammy : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e98e0e4cc96665bd67280b9bb428a6f6461bbc65","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-03-16 12:02:39.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/42/874242/3/check/fujitsu-eternus-dx-iscsi/46f4a40 : FAILURE in 59m 24s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/42/874242/3/check/fujitsu-eternus-dx-fc/5944c7d : FAILURE in 2m 20s","accounts_in_message":[],"_revision_number":3},{"id":"289d9264905e9f646a63d74019ce73b913c1950a","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-16 12:54:11.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/0bfe904be1aa4c4c91ff1298db85ac97 : POST_FAILURE in 42m 34s","accounts_in_message":[],"_revision_number":3},{"id":"4bc62b0899b342862d0fd077f9b5b8d5ee173ca3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-16 13:10:30.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/42/874242/3/check/tempest-dsvm-full-focal-py3/55fc7e0/ : FAILURE in 1h 34m 45s\n- devstack-plugin-ceph-dsvm-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/42/874242/3/check/devstack-plugin-ceph-dsvm-focal-py3/813a0d7/ : FAILURE in 2h 15m 26s","accounts_in_message":[],"_revision_number":3},{"id":"bdb4e54c11fb5a079172092e9605030974721d7a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-16 13:13:38.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/42/874242/3/check/iscsi-driver-alletra-6k/c7fbeb8/c7fbeb8.html : FAILURE in 1h 53m 04s","accounts_in_message":[],"_revision_number":3},{"id":"450ccde2e77bdd88a453fdf0e0869269691069d2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-16 19:33:38.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/42/874242/3/check/DellEMC_Unity_FC/2c0209f/DellEMC_Unity_FC/2511 : SUCCESS in 1h 33m 32s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/42/874242/3/check/DellEMC_Unity_iSCSI/fe906aa/DellEMC_Unity_iSCSI/2360 : SUCCESS in 2h 10m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"a709fe1484859415ac378d3b714efbf8034fdd95","author":{"_account_id":33609,"name":"XuQi","display_name":"Inori","email":"xuq.fnstxz@fujitsu.com","username":"inori"},"date":"2023-03-17 05:38:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"094b3706c5f18f8712e216128fed576c399e3884","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-03-17 07:18:42.000000000","message":"Patch Set 3:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/42/874242/3/check/fujitsu-eternus-dx-iscsi/5a36075 : FAILURE in 1h 37m 25s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/42/874242/3/check/fujitsu-eternus-dx-fc/f2a1305 : FAILURE in 1h 39m 26s","accounts_in_message":[],"_revision_number":3},{"id":"46252ff93b41ee6a96dd6ca767b3104de15934e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-03-23 12:31:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0f427c4f2945a7770fc71519b07a602018b40053","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-23 12:33:50.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"ca91e0d3053dfaf70341fded99a9c30c31ba894a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 12:44:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b06e1ea71c84b82b135ddf36067afca\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/69e050181e2348b0ba9696d0801e9777 : SUCCESS in 11m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/863d8ffe0f8340e4884b5c512e5281da : SUCCESS in 9m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"15c02e9a8284aa87049ceddce100171bf682e727","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-23 12:59:25.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/42/874242/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/34810a2/ : FAILURE in 11m 46s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/42/874242/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/2eeafc4/ : FAILURE in 11m 43s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/42/874242/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/5064d54/ : FAILURE in 12m 00s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/42/874242/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/32899b9/ : FAILURE in 11m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"019ba317a855919e76ba6e496ab649fcc74ebe37","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-23 13:03:02.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/874242 : FAILURE in 26m 45s","accounts_in_message":[],"_revision_number":4},{"id":"89a4dc67586ddbe0da62a596d30b6955cd54e971","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-23 13:56:43.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e9319207211c45729d83e10c607ca7ea : SUCCESS in 1h 22m 27s","accounts_in_message":[],"_revision_number":4},{"id":"31e40c913716b49b10bf2ad97b3437044f623feb","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-23 14:38:30.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/42/874242/4/check/iscsi-driver-alletra-6k/c5f3696/c5f3696.html : FAILURE in 2h 04m 20s","accounts_in_message":[],"_revision_number":4},{"id":"2c5b032cd4d7a2be4465ba210cd53929514812fe","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-03-23 14:39:02.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/42/874242/4/check/fujitsu-eternus-dx-iscsi/a34d3aa : SUCCESS in 2h 06m 00s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/42/874242/4/check/fujitsu-eternus-dx-fc/590f01e : FAILURE in 2m 30s","accounts_in_message":[],"_revision_number":4},{"id":"016165b0a61290257435b66a1ede9d885d7d1f3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 14:55:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/527ad342c6e748d789ea4901817f0995\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/727f529852f04511b7f1038c0583d69b : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1cc3da38da04130bb7236e0ecbcb252 : SUCCESS in 14m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/76d6db4c91c54e0688ec61ac4fd04ba1 : SUCCESS in 12m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7736e43cee764abfaf23c6970dff173f : SUCCESS in 12m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/620a1bb9262f4e838c157b88b9f570f8 : SUCCESS in 48m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d1dab8fef6fe42ba9fa24310ad2f2d90 : SUCCESS in 1h 21m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b56ae05e3c174020be352be56bbbfeea : SUCCESS in 1h 03m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/083715f018164cfc8f2a42bab3398200 : SUCCESS in 14m 47s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d3aaaf98a690489d9224250f69408fd6 : SUCCESS in 4m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c379c7a950f5498e863c358f993cd635 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7c2ad431f212441abaa30fd8807eda80 : SUCCESS in 6m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a3967545f03a44079b4abb1fe8617caf : SUCCESS in 5m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/ee6a9198fd0b44a986840dcc2df1befb : SUCCESS in 42m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0948b7266dfc4327bfe9381a7118850b : FAILURE in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4a2f9e476d3e44b9a12984de1f1f4c2a : SUCCESS in 1h 39m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/fa09ba6aada7451d81b9738da066adbd : FAILURE in 1h 34m 31s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dc748ea6b84a477fa17f8b7c6bbc9625 : SUCCESS in 2h 12m 26s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0e94c50a79b443939e15bec8d4ba599d : SUCCESS in 1h 47m 21s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4e1176c562d54c72bde9b7e9c20158c3 : SUCCESS in 1h 28m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/583c06d56eb945578028f9a7f57d088e : SUCCESS in 42m 20s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/928f47f653d2402e832dfe8be8eedbc7 : SUCCESS in 1h 33m 11s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7d89c0d08b994c54ad23d2eeeb4dec8c : SUCCESS in 2h 07m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f65321db38a4a539f3bc850597bfa34 : SUCCESS in 46m 31s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17b1bce8085142dd93f573629aa973f2 : SUCCESS in 1h 11m 48s","accounts_in_message":[],"_revision_number":4},{"id":"f6ba88ab9bc0b96be7bdf257b9ac69d5f30cff70","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-03-23 14:57:14.000000000","message":"Patch Set 4:\n\nBuild failed.","accounts_in_message":[],"_revision_number":4},{"id":"7f4f0790f19c3b683add2e245f1bea4e8c04215e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-03-23 15:03:39.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_03_2023_14_55_41\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"697aab2210b7bf1bc8b41d35dcac80d837972615","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-23 15:12:01.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/be8c81b047b844a199c2012064fe780c : SUCCESS in 1h 21m 43s","accounts_in_message":[],"_revision_number":4},{"id":"abe6b793632a4a0b94b4144a8d7ddd06d03bf397","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-23 16:19:47.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-42-874242-4/ns5_nfs : SUCCESS in 3h 20m 28s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-42-874242-4/ns5_iscsi : SUCCESS in 3h 24m 55s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"fe2bf5439818705b445b512eb7a3fc0779bf4392","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-03-23 16:24:25.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/42/874242/4/check/DellEMC_PowerFlex/07a5e4f/DellEMC_PowerFlex/1777 : SUCCESS in 1h 28m 28s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"aba795165e9d1861c9873aa196cc0a3269cec8ff","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-23 16:40:34.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/42/874242/4/check/DellEMC_Unity_FC/ff78978/DellEMC_Unity_FC/2557 : SUCCESS in 1h 43m 46s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/42/874242/4/check/DellEMC_Unity_iSCSI/1110c05/DellEMC_Unity_iSCSI/2404 : SUCCESS in 2h 11m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"07c01b3f2186b377c1828e2ebc08a34eab22db7e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-03-23 16:47:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/42/874242/4/upstream-check/cinder-tempest-plugin-ontap-fcp/a163587/ : SUCCESS in 1h 50m 14s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/42/874242/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/6658fd7/ : SUCCESS in 1h 15m 05s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/42/874242/4/upstream-check/cinder-tempest-plugin-ontap-nfs/78ad4af/ : SUCCESS in 1h 13m 05s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/42/874242/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/06f4627/ : SUCCESS in 44m 11s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"550313282ddb996cd9bf5ed9474509b13ed27085","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2023-03-23 16:47:19.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/42/874242/4/check/DellEMC_SC_FC/79ca2e5/DellEMC_SC_FC/1243 : SUCCESS in 1h 06m 21s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/42/874242/4/check/DellEMC_SC_iSCSI/d15837e/DellEMC_SC_iSCSI/1223 : SUCCESS in 1h 51m 25s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":4},{"id":"2e4693761b181b28a85e49e38acb6197b884365f","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-03-23 18:43:28.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/874242-4/HPMSA-iSCSI/ : SUCCESS in 1:40:18\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/874242-4/HPMSA-FC/ : FAILURE in 1:16:52","accounts_in_message":[],"_revision_number":4},{"id":"8e4526e04172d67d840acf77fadda3efbfadcdec","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-03-23 19:04:17.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/874242-4/Seagate-iSCSI/ : SUCCESS in 1:46:54\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/874242-4/Seagate-FC/ : FAILURE in 1:30:56","accounts_in_message":[],"_revision_number":4},{"id":"ee0b51b07efdbd3ca73021e2b46f4fb8ef66e61b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-24 04:00:28.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder iscsi-driver-primera-cinder : RETRY_LIMIT in 0s","accounts_in_message":[],"_revision_number":4},{"id":"785ff32baf974a114aa810318d6d50b828fcf52b","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-03-25 18:24:20.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/42/874242/4/20230326003511/NEC-V-ISCSI : SUCCESS in 2h 17m 47s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/42/874242/4/20230326003511/NEC-V-FC : SUCCESS in 2h 49m 07s","accounts_in_message":[],"_revision_number":4},{"id":"c31a6081f4608706fb624d89ffcc4b9dc3ec35dc","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-03-25 21:35:08.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/42/874242/4/ : SUCCESS in 1h 20m 27s","accounts_in_message":[],"_revision_number":4},{"id":"c3639b8d2f0fcabbce79f9f0ae6d5391d2ba2801","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-22 20:08:24.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/42/874242/4/check/iscsi-driver-nimble/90fcc74/90fcc74.html : FAILURE in 36m 14s","accounts_in_message":[],"_revision_number":4},{"id":"68f342cfa2752d659b4045706b18cbc6864e8f77","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-04-29 23:51:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/42/874242/4/20230430052759/NEC-ISCSI : SUCCESS in 3h 16m 41s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/42/874242/4/20230430052759/NEC-FC : SUCCESS in 3h 23m 07s","accounts_in_message":[],"_revision_number":4},{"id":"84082bc37b1cdbf3f5cfdaf5251f4ee79e0c6c53","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-05-23 09:43:48.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"131e060da566e83844fa0ad7989849f54697891b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-05-23 10:02:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"50a21bd41e5d53d6585d5e8a403de7f85cc73ca4","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-06-26 07:55:02.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"baaf51dac4aaaf8ba3f736dc469fbf21f3d9b8fe","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-26 12:30:30.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-42-874242-4/FC/ : UNSTABLE in 48m 42s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-42-874242-4/iSCSI/ : FAILURE in 4h 2m 54s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"32869fe073f8285a02f3b813138f8e0d4b339f4a","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-27 01:00:08.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/42/874242/4/fc : SUCCESS in 0h 49m 01s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/42/874242/4/iscsi : SUCCESS in 0h 51m 35s","accounts_in_message":[],"_revision_number":4},{"id":"99305511559c48830fad3569aab710f1b704ea30","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-30 17:00:29.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-42-874242-4/FC/ : UNSTABLE in 24m 7s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-42-874242-4/iSCSI/ : FAILURE in 4h 22m 40s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"e36d39e0a74f198fdab39c4a4896c6dd7a968ef4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-06-30 18:34:48.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c8407245ddb1aea44bcbe4945ee0ffe765700cce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 18:35:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9b89493015da5e1d60c272334f0f56267f9a6b3d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-30 23:40:05.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-42-874242-4/FC/ : FAILURE in 5h 44m 35s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-42-874242-4/iSCSI/ : FAILURE in 4h 16m 47s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"d6477418f041ed2c4062b51436adf61c9a7a4fcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-01 03:05:39.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b7015ea585a4c79a4b50421249b035e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a35e5ac0ea404b3bad2398061793232a : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2c704398a4440ef9031cba889f4f4ca : SUCCESS in 11m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3db8fb88d0ef4d29ab7e9002158635f9 : SUCCESS in 13m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9f617e02d2c4a20b3d28c1a43b8fdfe : SUCCESS in 12m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7678b575c0764151995cd0ee2d5c2028 : SUCCESS in 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/af16c58f3647421bb379bf9abbbb84e0 : SUCCESS in 52m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e93fd1a4ff51458d9fef9101942223b9 : FAILURE in 1h 07m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6038a826f21545cf8baf1f0130fe6674 : SUCCESS in 48m 23s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/fb3592282ead4a66903a2669927dbb1f : SUCCESS in 1h 15m 18s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a43c6835a8e34fe5a2ca11b9701640a2 : FAILURE in 2h 10m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b745e59929664bb5ba2a3830f5c6d03b : SUCCESS in 40m 07s","accounts_in_message":[],"_revision_number":4},{"id":"38c1a170b8b2b1188489d6b7448d2786f9d1ab3d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-03 11:15:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1772c8a563e502e16bcfac827ec25afd48db5bae","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-03 11:16:27.000000000","message":"Patch Set 4:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3683104c8d08a941218d70913ab81d79ce819b12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 11:31:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a9bac8f81ea49acaa912eab866a0757\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e851fe91be5a41c1a8855c7cf7d086d2 : SUCCESS in 12m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/327a95930b794b869cf769713114fc44 : SUCCESS in 11m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/be7ed2413d66416384e4e1ec577cb288 : SUCCESS in 12m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"be680e888f0fd3b9cff99d9b00fae587dad0065e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-03 12:54:41.000000000","message":"Patch Set 4:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/e71b1b61dbb64fc9a3f53b23893e3502\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0559f88136a846b8971286899290fdc0 : SUCCESS in 1h 16m 57s","accounts_in_message":[],"_revision_number":4},{"id":"f662108f947051c247e8249c38cbc65014e75d7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 14:26:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1ff42f7bfe84808a1ec0f9edab10362\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf1d6114b33240bc896edd918168a6d1 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be5850ef1f8a48cf9538956a9a3f5bbf : SUCCESS in 13m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e9a93062592486fbf93149662af4791 : SUCCESS in 9m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/529ab4627a2f454495402a255e8d6a9f : SUCCESS in 11m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/925321a17d3a437f80c514b62c490328 : SUCCESS in 14m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/062f64c9c691421a9bda0395a334cf81 : SUCCESS in 1h 37m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9ebbbf2f0a5d40198a1ba22b36cd2b19 : SUCCESS in 1h 40m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c46c677692eb4533870f581d56a6de48 : SUCCESS in 1h 01m 04s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9eb2649693a1425b88ea09447d050da0 : SUCCESS in 15m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/88913a6292a248b4bdd0c193cc7a3cdb : SUCCESS in 6m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/502b99fe0ba24671abd3a56978652031 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1e776bc1456149669f0a86f8d509101d : SUCCESS in 6m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2d1ff8e5218f4723a48cc7409db82feb : SUCCESS in 5m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/32ad594327224fcbb173277edeccd308 : SUCCESS in 1h 03m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/11cb23dfa7b54563a92acd99ee07efbd : SUCCESS in 6m 08s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/54f520782467430a82868466201441a4 : SUCCESS in 2h 24m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b59e95aa6fed465190ff399ddfb1d2c0 : SUCCESS in 1h 45m 05s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/df9291f47dca43c582ca5c647836bb3a : SUCCESS in 2h 33m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/2c7d426da3634a81b339893109474f51 : SUCCESS in 2h 40m 40s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/7762d440bdf2403fbaa3528033c5c597 : SUCCESS in 1h 30m 33s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9462a5ffa777436fa08cd11a46d43724 : SUCCESS in 1h 55m 46s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a04fc2ab332b46109de98d6caf462333 : FAILURE in 1h 15m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c78b3cfb275949b294d07513ea97a7c4 : SUCCESS in 1h 40m 01s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/19fc83cc6aa144f297b61d0bb7d59533 : SUCCESS in 1h 50m 26s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c25f53e8c86a41d0afb6335a2013b2fd : SUCCESS in 3h 02m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79dd2e0fbda64e228cfb74ba970b2260 : SUCCESS in 1h 08m 32s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/51c2bfb9048c4983ace35800acbb08ee : SUCCESS in 11m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e6cc85952d0404952be57f89c0ccecac2c5db168","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 14:27:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9d1277cafd9cbda3cd16657e076b2aec11781b77","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-07-03 14:33:26.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/03_07_2023_14_26_21\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"4e701fbca11a636ec4dd53da7cf3530a033dc2a2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-03 15:14:41.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/42/874242/4/upstream-check/cinder-tempest-plugin-ontap-fcp/85ca1a7/ : RETRY_LIMIT in 7m 02s (non-voting)\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/42/874242/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/7e33b4d/ : RETRY_LIMIT in 7m 33s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/42/874242/4/upstream-check/cinder-tempest-plugin-ontap-nfs/e4b59fb/ : SUCCESS in 47m 52s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/42/874242/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/f96f8f8/ : FAILURE in 46m 34s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"e3f1e724a69f0b1d95c8907fc11a4c8cab2e0914","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-07-03 15:47:20.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/42/874242/4/check/DellEMC_PowerStore_iSCSI/747027e/DellEMC_PowerStore_iSCSI/137 : SUCCESS in 50m 56s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/42/874242/4/check/DellEMC_PowerStore_FC/e2541b2/DellEMC_PowerStore_FC/145 : SUCCESS in 1h 20m 46s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/42/874242/4/check/DellEMC_PowerStore_NFS/c80171c/DellEMC_PowerStore_NFS/136 : SUCCESS in 54m 15s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/42/874242/4/check/DellEMC_PowerStore_NVMeTCP/c9a6401/DellEMC_PowerStore_NVMeTCP/137 : SUCCESS in 1h 10m 36s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"0b889ed9adac6ecef2adbd6aa96bf98d1ef07ed9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-03 15:58:54.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/42/874242/4/check/DellEMC_PowerFlex/a67484d/DellEMC_PowerFlex/252 : SUCCESS in 1h 32m 24s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"c504e48942205a9d38982f32f2c67b5f78448a17","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-03 16:45:57.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/42/874242/4/check/DellEMC_Unity_FC/0dda8c2/DellEMC_Unity_FC/370 : FAILURE in 1h 54m 02s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/42/874242/4/check/DellEMC_Unity_iSCSI/efea389/DellEMC_Unity_iSCSI/372 : FAILURE in 1h 23m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"b468e8cc72db796fe3ca325cd26b91555c7d3332","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-07-03 19:24:54.000000000","message":"Patch Set 4:\n\nBuild failed.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/874242-4/HPMSA-iSCSI/ : FAILURE in 1:53:43\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/874242-4/HPMSA-FC/ : FAILURE in 1:31:33","accounts_in_message":[],"_revision_number":4},{"id":"12fb5b6ec6805fe5a931ad68ccaa45e1c960484b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 19:31:33.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dcd310bb2f84a4e8d730e7b58a928cd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad411c137f414ead9fd990b13fd2dae5 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ef46df87e0741b9924ce3d8db4ec74c : SUCCESS in 11m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f3397a871a54552b27f9194f4cdf1cc : SUCCESS in 12m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6448da2d9df4befa1b6d96f15103874 : SUCCESS in 10m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a39fcf61f3ca41318a13503489d57d49 : SUCCESS in 11m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/ef020b49a5404860ba5d24345cab4088 : SUCCESS in 51m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/23d8b834059c47ecb779f0eac47e4a46 : SUCCESS in 1h 46m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f64e443641e0424988f6dac70878b85c : SUCCESS in 58m 24s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1ecdd90e9530401ba8b5c24ab4f71d98 : SUCCESS in 1h 13m 16s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5ccda8e6c1014f5cbbe8a84738db0751 : SUCCESS in 2h 01m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd6ab71283db42e78513e552f4691c78 : SUCCESS in 50m 55s","accounts_in_message":[],"_revision_number":4},{"id":"8419d62a2426973f4a33b510a56cb49781a5780e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 19:31:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"93388d7e7c5ac8934e7836725bb9aa9bb7ce69d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 19:32:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/023d1b1b97d74241b5e56f372916457d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62aa42b5b9db409fba193980fbf69483 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":4},{"id":"42f43c7f6c5dc649aad9d348a207c07686a329b4","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-07-03 20:00:31.000000000","message":"Patch Set 4:\n\nBuild failed.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/874242-4/PVME-iSCSI/ : FAILURE in 2:09:50\n- PVME-FC https://os-logs.seagate.com/dell-ci/874242-4/PVME-FC/ : FAILURE in 1:40:00","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"171ff11e905a2103443c547d28f84f98a651491c","revisions":{"ac97230ef3e146ead4cce1a5a117c2f700e8b2aa":{"kind":"REWORK","_number":1,"created":"2023-02-17 12:57:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/42/874242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/874242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/874242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/874242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/874242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/874242/1"}}},"commit":{"parents":[{"commit":"1e69a5be77927f266cc48786634ddc456b0f0ad0","subject":"db: Remove legacy migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1e69a5be77927f266cc48786634ddc456b0f0ad0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-17 12:42:53.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-17 12:57:06.000000000","tz":0},"subject":"db: Remove the legacy \u0027migration_version\u0027 table","message":"db: Remove the legacy \u0027migration_version\u0027 table\n\nThis hasn\u0027t been necessary since Xena.\n\nChange-Id: Ia1b92341fc44fa903c238b2cb82496b77012a367\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac97230ef3e146ead4cce1a5a117c2f700e8b2aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ac97230ef3e146ead4cce1a5a117c2f700e8b2aa"}]},"branch":"refs/heads/master"},"036800f47ab37126b731a0297cdff579d6af5f89":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-17 15:44:28.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/42/874242/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/874242/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/874242/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/874242/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/874242/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/874242/2"}}},"commit":{"parents":[{"commit":"63a83c3ffa09785ea64f7519ca47d6e32da40a1c","subject":"db: Remove legacy migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63a83c3ffa09785ea64f7519ca47d6e32da40a1c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-17 12:42:53.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-17 15:44:16.000000000","tz":0},"subject":"db: Remove the legacy \u0027migration_version\u0027 table","message":"db: Remove the legacy \u0027migration_version\u0027 table\n\nThis hasn\u0027t been necessary since Xena.\n\nChange-Id: Ia1b92341fc44fa903c238b2cb82496b77012a367\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/036800f47ab37126b731a0297cdff579d6af5f89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/036800f47ab37126b731a0297cdff579d6af5f89"}]},"branch":"refs/heads/master"},"a445aa1db495d217596c3567a768c9f08dbf41f5":{"kind":"REWORK","_number":3,"created":"2023-03-16 09:18:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/42/874242/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/874242/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/874242/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/874242/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/874242/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/874242/3"}}},"commit":{"parents":[{"commit":"784e92afbe817c47eb01e555c343854a07624945","subject":"db: Remove legacy migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/784e92afbe817c47eb01e555c343854a07624945"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-17 12:42:53.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-03-16 09:18:26.000000000","tz":0},"subject":"db: Remove the legacy \u0027migration_version\u0027 table","message":"db: Remove the legacy \u0027migration_version\u0027 table\n\nThis hasn\u0027t been necessary since Xena.\n\nChange-Id: Ia1b92341fc44fa903c238b2cb82496b77012a367\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a445aa1db495d217596c3567a768c9f08dbf41f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a445aa1db495d217596c3567a768c9f08dbf41f5"}]},"branch":"refs/heads/master"},"171ff11e905a2103443c547d28f84f98a651491c":{"kind":"REWORK","_number":4,"created":"2023-03-23 12:31:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/42/874242/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/42/874242/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/874242/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/874242/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/42/874242/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/42/874242/4"}}},"commit":{"parents":[{"commit":"e9dccb93be059176fec19b9a021672cbcc2f8414","subject":"db: Remove legacy migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e9dccb93be059176fec19b9a021672cbcc2f8414"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-17 12:42:53.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-03-23 12:24:52.000000000","tz":0},"subject":"db: Remove the legacy \u0027migration_version\u0027 table","message":"db: Remove the legacy \u0027migration_version\u0027 table\n\nThis hasn\u0027t been necessary since Xena.\n\nChange-Id: Ia1b92341fc44fa903c238b2cb82496b77012a367\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/171ff11e905a2103443c547d28f84f98a651491c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/171ff11e905a2103443c547d28f84f98a651491c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
