)]}'
{"id":"openstack%2Fcinder~874690","triplet_id":"openstack%2Fcinder~master~I4abfb6de58340c5ea70e7a796326f5d2089c80eb","project":"openstack/cinder","branch":"master","topic":"pure-addressing-mode","attention_set":{},"removed_from_attention_set":{"13425":{"account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"last_update":"2023-09-16 02:26:34.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-09-15 23:32:54.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"35075":{"account":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"last_update":"2023-09-16 02:26:34.000000000","reason":"Change was submitted"},"29268":{"account":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"last_update":"2023-09-15 17:21:10.000000000","reason":"\u003cGERRIT_ACCOUNT_13425\u003e replied on the change","reason_account":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-09-16 02:26:34.000000000","reason":"Change was submitted"},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2023-09-16 02:26:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4abfb6de58340c5ea70e7a796326f5d2089c80eb","subject":"Pure: Report SAM-2 addressing mode for LUNs","status":"MERGED","created":"2023-02-21 19:23:57.000000000","updated":"2023-09-20 16:19:30.000000000","submitted":"2023-09-16 02:26:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"874690-pure-addressing-mode","meta_rev_id":"97c182d941dc02394c5f71ee485a47c49d1867eb","_number":874690,"virtual_id_number":874690,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-16 02:26:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":2,"date":"2023-09-15 23:32:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2023-09-15 23:32:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-21 19:24:09.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-21 19:24:16.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-21 19:48:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-21 19:49:45.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-21 19:51:40.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-02-21 22:22:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-21 22:28:17.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-21 23:23:46.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-02-22 01:52:20.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-02-22 02:15:24.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-22 02:25:39.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-22 06:52:19.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-02-22 08:44:10.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-22 14:44:13.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"CC"},{"updated":"2023-02-25 01:13:11.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-03-01 21:03:04.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-07 14:04:54.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2023-03-10 10:54:14.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-03-10 13:13:10.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-03-10 14:22:16.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-03-10 14:32:09.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-03-10 14:56:47.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2023-03-10 22:25:31.000000000","updated_by":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"reviewer":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"state":"CC"},{"updated":"2023-03-11 23:47:56.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-03-12 19:41:41.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-03-13 06:59:59.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2023-07-27 16:32:04.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-07-27 20:00:12.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-09-14 14:17:55.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-09-15 14:45:14.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-09-15 14:52:11.000000000","updated_by":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"reviewer":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"state":"REVIEWER"},{"updated":"2023-09-15 20:26:13.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-09-17 13:09:22.000000000","updated_by":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"reviewer":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"state":"CC"},{"updated":"2023-09-20 04:22:32.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"e7170f14d05dea239c40669e6d542e956ada78eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2023-02-21 19:23:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a616992d65d23ad3aa1845f9378f33f8db73c068","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2023-02-21 19:24:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2eac6c865af5ef46ead1c6f10c092536b176a3f5","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-02-21 19:24:16.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"45c21eba5278836125c1bc820e8bbc6d1bf05fc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 19:48:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a84ce76e68de42f990de2ffc3f95d7f8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4282c9efc9fa49239851baf35db9bee8 : SUCCESS in 20m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1efc2056768849aea5e19e68544b7b96 : SUCCESS in 10m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3643fd5ae10fb86a433ef4c9d641c170fabc15e","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-21 19:49:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder https://github.com/hpe-storage/hpe-primera-logs/tree/main/new_logs/874690 : FAILURE in 23m 28s","accounts_in_message":[],"_revision_number":1},{"id":"5c2dd33fdd2590124e94161e824846b433185aa3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-21 19:51:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/90/874690/1/check/iscsi-driver-alletra-6k/8e96cea/8e96cea.html : FAILURE in 26m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d151ceb183b303aca03a2a40538dc7a28d402ad6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-02-21 19:52:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/874690 : FAILURE in 26m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1af5af9848718e7a045a4b3a01475cfcc20181e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 22:22:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff8e7c5da2524d14a5d693cb96b76ccd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a05faee52e84329a85e62492dfec822 : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c82983a79cc45e3a19e6482c84f9b64 : SUCCESS in 11m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0b699f88ea847bd910293a36bdb79bc : SUCCESS in 10m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/858fe7a7c86d4ccaa85e31996a3a5824 : SUCCESS in 13m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/5ffeacf2da644b82b377c68a7319393c : SUCCESS in 1h 02m 09s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5ee1763de3384d46a03bfb0ab9e85d22 : SUCCESS in 1h 01m 40s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/93ec32e651e3452fa225769b3accce75 : SUCCESS in 1h 43m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ba56b1dbb9a47639801c14a8231f9a9 : SUCCESS in 1h 05m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2ee5d60a0464d9baf740113abcddf76 : SUCCESS in 9m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/bfbda43c3a024778bd38f0af36e65bfd : SUCCESS in 15m 20s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/913462e8bac24602b2bcff00399cbc5e : SUCCESS in 6m 49s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/811d501e5e354af3b3392eec4fd7ec37 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/32d4fed1210b47119b8b827101152b14 : SUCCESS in 6m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/33b92606cd3c41b7a503d9673c7f0540 : SUCCESS in 6m 49s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/94434ca3e73543bdbc52624698108c2b : SUCCESS in 46m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ba6037755e1d4c159aedf5aa12a60251 : FAILURE in 6m 49s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3d1c902a70eb4436ae4564398af62634 : TIMED_OUT in 2h 04m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/5722991fe7fa4b5887fb9ee56e70f000 : FAILURE in 2h 44m 44s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fab4cb27993f4cf8bbe0276bb126052a : SUCCESS in 2h 56m 44s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f2df964fa54d4add8af7de0d40b72992 : SUCCESS in 1h 02m 55s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/dd932835746b4f58a165ec15ccdad6a8 : SUCCESS in 1h 34m 16s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/54499460db09489f820558b7c666246c : SUCCESS in 33m 49s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/4cdb84b067d94947ab4a752f5297dac8 : SUCCESS in 1h 27m 13s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/afe3686cd60c4dba9c44212b6bc60b88 : FAILURE in 2h 26m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92a88a60816a48d59c4ccde1340a07f5 : FAILURE in 50m 44s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a2be757b4fb4b4b8e67f7e419793308 : SUCCESS in 1h 21m 06s","accounts_in_message":[],"_revision_number":1},{"id":"55db3b0b44cdec9ee42f3c20bbc6b019102b711f","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-02-21 22:28:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/90/874690/1/ : FAILURE in 1h 02m 19s","accounts_in_message":[],"_revision_number":1},{"id":"eb05abd91553bb9da0b2ef4fe8eadc5af07f5049","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-21 23:23:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2db913a/ : TIMED_OUT in 3h 08m 07s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/d90dbce/ : FAILURE in 2h 26m 59s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/28a6f0e/ : SUCCESS in 1h 52m 21s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/e36c5ee/ : FAILURE in 2h 24m 31s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy pure-devstack-cinder-tempest-nvme-roce-aio-jammy : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5bb9a190b79cf18c5a05364413a59c03b78b6275","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-02-22 01:52:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- run-devstack-cinder https://ec2-3-95-28-238.compute-1.amazonaws.com/zuul/t/local/build/b9e72217152f4d8391b37a1a188777ef : SUCCESS in 1h 13m 58s","accounts_in_message":[],"_revision_number":1},{"id":"47899359bf4c43ff3ed4369454ed3dbed5c16577","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-22 02:15:24.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/e022eb797d934c6497cb35c6481117f2 : SUCCESS in 2h 02m 50s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/a6d3f4a8cf5540b6b3aafbc1dfb5d7f1 : FAILURE in 1h 50m 56s","accounts_in_message":[],"_revision_number":1},{"id":"bce35da86744e5457c927a5a369f227f7fb0a6df","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-02-22 02:25:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8658b97923344d128b038c2028b0f365 : SUCCESS in 1h 11m 36s","accounts_in_message":[],"_revision_number":1},{"id":"43c670fc88fb6293a1021f590c3daf0ded3ad5ef","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-22 06:52:19.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/90/874690/1/check/DellEMC_Unity_FC/c6c5d00/DellEMC_Unity_FC/2338 : FAILURE in 2h 54m 33s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/90/874690/1/check/DellEMC_Unity_iSCSI/1d9647c/DellEMC_Unity_iSCSI/2192 : FAILURE in 2h 01m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"3cadc01ed96f19af5d67b49202a3f6ae8ab64e94","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-22 08:44:10.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-874690-1/ns5_nfs : FAILURE in 1h 34m 33s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-874690-1/ns5_iscsi : FAILURE in 13h 18m 24s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"b330df1926112d9e28824806062c90ade5902de5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-02-22 10:11:45.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"98fb8889121025446d72e33a133d9236b96485e9","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2023-02-22 10:12:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81369799705f0efb91eadd2fe6495e835b6e7d04","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-22 13:41:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/105172a/ : SUCCESS in 1h 50m 13s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/e6e36d5/ : FAILURE in 2h 18m 06s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio pure-devstack-cinder-tempest-nvme-roce-aio : NODE_FAILURE in 0s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/bf27eb2/ : SUCCESS in 2h 06m 29s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/97d9dd2/ : FAILURE in 2h 19m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7157324c0016978e506ea2d46535e3c519c2cf6f","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-02-22 14:44:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7d79d2ac243b5d160664dd307183559352a9733f","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-02-22 14:45:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eca74a8121b6145c29c1d363675b9074d20d2374","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-22 17:19:17.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/8edea78/ : SUCCESS in 1h 44m 23s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/6c19632/ : FAILURE in 2h 32m 33s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/5088368/ : SUCCESS in 1h 33m 30s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/c15761a/ : SUCCESS in 1h 45m 00s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/238d019/ : SUCCESS in 2h 03m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f8a5ed3a274a7ee0f885544c8c76151d275af98a","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-02-22 21:21:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a73cb95ec40abd214df4b83d5d590e16f270f2c0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-22 23:16:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/9efe236/ : SUCCESS in 1h 43m 53s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/9592662/ : FAILURE in 1h 52m 18s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/36fb8cf/ : SUCCESS in 1h 36m 41s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/f005fdd/ : SUCCESS in 1h 49m 32s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/14979cc/ : SUCCESS in 1h 48m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9487fe544d23c46877fd30ce754c1ebaf7766305","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-02-25 01:13:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/874690/1/20230225073558/NEC-V-ISCSI : SUCCESS in 2h 06m 58s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/874690/1/20230225073558/NEC-V-FC : SUCCESS in 2h 37m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c1ea9f2ac938e4dc0e11b5526b95666cb449fe46","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-03-01 21:03:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/874690/1/20230302022842/NEC-FC : SUCCESS in 3h 30m 40s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/874690/1/20230302022842/NEC-ISCSI : SUCCESS in 3h 34m 19s","accounts_in_message":[],"_revision_number":1},{"id":"5e942cd061e86af784292ab59dd9b0f2217d5a7a","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-03-07 14:02:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c0d4d93f660cd41fe85b6a1306af2defd26cc1f","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2023-03-07 14:02:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6b772c73c2cc71df6f5f20c8112094f7e8c1341","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-03-07 14:02:23.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"c3d23ebbdeac97417e6462bb06fda808ea12643e","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-03-07 14:04:54.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e0349ef4d66a236d03c4655701035fb20ff7977","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 14:29:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f47bd489bd434117be0369a5313948f3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/264cdce44001428ea339bd03c17f7ea0 : SUCCESS in 22m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/026be7b3f0d5498f9920aba61e6cdb6f : FAILURE in 25m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06a67cabec1e45ecc096ec1e8d012711099e1fc2","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-07 14:34:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/83e6615a573245a6ab30f1bdb23e3ffc : POST_FAILURE in 26m 51s","accounts_in_message":[],"_revision_number":1},{"id":"3c45eee5b22f6ac3f26bea67d96fcab64c166014","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-07 15:11:41.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/98b55dd2946e4af7acc72ab03040eee6 : SUCCESS in 1h 07m 37s","accounts_in_message":[],"_revision_number":1},{"id":"15a437ecc7993f6d80d56f49cbc27fcb8c78028e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-07 16:25:20.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/92f218b/ : SUCCESS in 1h 39m 21s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/9a2492c/ : FAILURE in 2h 01m 30s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/08d8b40/ : FAILURE in 2h 08m 33s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/1404d65/ : SUCCESS in 1h 45m 34s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/90/874690/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/bacb9ee/ : SUCCESS in 1h 50m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a12e19b24f96307d5b479995c4eef42b8d70534","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 16:28:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/15f3d1cf5caf4d84b66a62c31ebed69b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db0e8a4cddb24511a6f7aae4c472243e : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d812655c311246e081109a574d7f3c94 : SUCCESS in 11m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98fd7bf6e4e24b21840f183c00e54976 : SUCCESS in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c786428de404d828b78b84ef9c25613 : SUCCESS in 12m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/e13c8338ddc948f4a4778d5a29815ea4 : SUCCESS in 50m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/adf6fa851fe540238161e8dc227c122c : SUCCESS in 53m 24s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/54b7c9dd53c140c9a0fd45792695c3f7 : SUCCESS in 1h 24m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1780ce4df3554bcebcf1078cb355184b : SUCCESS in 1h 00m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a86573a4707743678b64136fff499849 : SUCCESS in 8m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/a084a50957594a5b9d6349db95e721a1 : SUCCESS in 14m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/474ee9c03d264a66bb2b9b369ea98737 : SUCCESS in 5m 34s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/59075cda70fa4932900af6a0386167bf : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6f0ee7d9e669400d822f53a2e2c2c9a2 : SUCCESS in 5m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ee23db17e2694495bcdda4a1b04df12d : SUCCESS in 5m 47s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/116eea9f67404531b7f4d39a536deb26 : SUCCESS in 1h 36m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ba6876a4f36343fca548673c32c29564 : FAILURE in 5m 23s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/bf499786a3fe42d38280f0da2d2d22e0 : SUCCESS in 1h 00m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8d4d3a49749c4de0b92765c0b18c230a : FAILURE in 1h 18m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/96cebbd2613f4b5c83a46cbd70f8356c : FAILURE in 1h 45m 18s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cfef69bd1fd34faca82b5a866ea8dd09 : SUCCESS in 1h 20m 32s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a9a334e19a8e449bbec9473e3a77f4aa : SUCCESS in 1h 25m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2f827da4a3504b7d9c3219578a521547 : SUCCESS in 42m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/30c64b5ce79a42e6abf5a4a9f1799e42 : SUCCESS in 1h 03m 10s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/457dc9cd1d63443685929c07fef948b5 : SUCCESS in 2h 17m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8564cd1228764e94a214e3e9947e6d72 : SUCCESS in 56m 28s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d65438253c3d48c2b33a3b28ff8c7190 : SUCCESS in 1h 24m 40s","accounts_in_message":[],"_revision_number":1},{"id":"3c628dc414a7f3dd2386a25328f79583b3f8f4ec","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-07 22:58:33.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/b7d8941f19c74e588380e26555aee573 : SUCCESS in 1h 44m 14s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/0bfd7f5a60e747bd921d25f7b6a3a285 : SUCCESS in 2h 22m 59s","accounts_in_message":[],"_revision_number":1},{"id":"3ae9a93270fd56cc64185c068d28c37be61cec85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2023-03-10 10:52:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"026ffd684aef70d7baeb17438ff35ba5212f77dc","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2023-03-10 10:53:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-zadara-driver dsvm-tempest-zadara-driver : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8a1ac97ee3cba3728d8ad5a031476d82f96e7576","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-03-10 10:53:14.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"84c5844d8ea6a2c52e32c2fa5b96f666d912f90f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-10 10:54:14.000000000","message":"Patch Set 2:\n\nBuild Started http://openstack-ci.spb.yadro.com:8080/job/openstack-gerrit-trigger-fc/75/ (1/2)","accounts_in_message":[],"_revision_number":2},{"id":"1376d977f3eaaa33a962b8745aa098c989eee46d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-10 11:18:53.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder finger://openstack-ci-fc/4ed83d07d7ef456c8992b502b2fa80f4 : RETRY_LIMIT in 1m 16s","accounts_in_message":[],"_revision_number":2},{"id":"ca18116eed082ee17a42012674f250aa82ffd223","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 11:20:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb0bfdd5506d4addb5bef07a8f9c1192\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad7b718d198a491eb60e63129db69ace : SUCCESS in 21m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d625cf89b021481d9a33c23d20dfda3e : FAILURE in 24m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d253c6445f1f3993973e5632c2c0832bde73dce7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-10 11:23:34.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/874690 : FAILURE in 26m 48s","accounts_in_message":[],"_revision_number":2},{"id":"f619f8202a9dc532433bafe2a4f4c54066bb78b9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-10 12:04:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2c7f1513afd248a49b806051a56b268a : SUCCESS in 1h 09m 54s","accounts_in_message":[],"_revision_number":2},{"id":"8b3dcf0af9f73bef0b520315ff23415432ca58aa","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-03-10 12:09:19.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/90/874690/2/ : FAILURE in 1h 06m 30s","accounts_in_message":[],"_revision_number":2},{"id":"609be0bec7f4dc4b9b02b5c924afde14b02e6853","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-10 12:40:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/90/874690/2/check/iscsi-driver-alletra-6k/d27ba99/d27ba99.html : FAILURE in 1h 45m 36s","accounts_in_message":[],"_revision_number":2},{"id":"77aad540b591fd60ad46afaf54c73fcd41fdda8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 13:04:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1268991e632a45f5b0fc85e70156942e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/818aa25f48704b9c91e05a01e94eccd8 : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cc411276c76456aacb15aa00342abff : SUCCESS in 10m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8645b4a263c444298d4aa12fda986d6 : SUCCESS in 11m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c81b696b333048939beed09f65e64967 : SUCCESS in 12m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/1c3612fe188342a4947bedef9cabe1fa : SUCCESS in 1h 00m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/981c7942f05b43808e83002050bb1ce1 : SUCCESS in 1h 04m 47s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6e561df680c6454bb9894bfc3a9728d1 : SUCCESS in 1h 31m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4fa5bffca4ef48598b38ff66793a56e6 : SUCCESS in 59m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8442ff54e761470c9091cf84aa8a9032 : SUCCESS in 8m 21s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/28c8ebecb0ee4e10b5ed005258bae51a : SUCCESS in 14m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/de14feb17c5f4f098e1c7162ef805f5e : SUCCESS in 5m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/82f7b14ddf5149dab95a13834a073775 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8a7ff148a635415191fefbd6282f90ad : SUCCESS in 5m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3cb34a49cabb45e99dd7205d6470fffc : SUCCESS in 6m 46s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/de16479a4a974e79b122aedacdac85f2 : SUCCESS in 57m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4f8b734c565a465380db4f79e1656fd5 : FAILURE in 6m 13s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/70e6a129a1f74701808257a88299071c : SUCCESS in 1h 25m 11s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/560c07a585964c96b8867cd15503edba : FAILURE in 1h 23m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1ff18f01c1c94764a3503c7419128550 : SUCCESS in 1h 35m 11s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4924b994dff84c56ac7a3764032b7c34 : SUCCESS in 1h 23m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/24665a4ef34f4503b5bf38fb6c87a565 : SUCCESS in 1h 29m 04s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d4e9ead4f95c4a53848766a733096378 : SUCCESS in 36m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d5465edec30148b9864513eae2244956 : SUCCESS in 1h 19m 08s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ae8c7a75f1824dbaba4641121a21702f : SUCCESS in 2h 06m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06c852640d774182b444e61aa346c253 : SUCCESS in 1h 02m 31s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6c62bf917e74c1fb29a03dd2be3f43e : SUCCESS in 1h 28m 36s","accounts_in_message":[],"_revision_number":2},{"id":"88559c9b69cf0e91e5098046219e0e53e95bc36a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-03-10 13:13:10.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/10_03_2023_13_04_45\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"89504f5847b882cda0b228673ea413675a633eb7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-10 13:18:05.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/90/874690/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/2ac76d7/ : SUCCESS in 1h 35m 40s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/90/874690/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/7c54608/ : FAILURE in 2h 22m 56s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/90/874690/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/8771036/ : SUCCESS in 1h 39m 41s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/90/874690/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/aa63e75/ : SUCCESS in 1h 53m 32s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/90/874690/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/f192acd/ : SUCCESS in 1h 59m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5cb1e698cf2023d41d3064b828adb8d1187b5396","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-03-10 13:24:42.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7867064f46eac6ae9ac16809a4110a3ff02eb24b","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-10 13:25:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/1b6d3f1b311347f1972e484049dd8695 : SUCCESS in 1h 22m 41s","accounts_in_message":[],"_revision_number":2},{"id":"f0c9b6c9c8db545e936110a2dd1bb130901ef7c6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-10 13:26:09.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/944e9b6317e5409ba914ccb5287a5179 : SUCCESS in 1h 56m 22s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/ec2167076cdc4643ac7fbcd07ce6dcff : FAILURE in 2h 13m 00s","accounts_in_message":[],"_revision_number":2},{"id":"36d593bbb5b788859b27d7a67243f3bd6f1e6957","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-03-10 14:22:16.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/874690/2/upstream-check/cinder-tempest-plugin-ontap-fcp/a5b36fe/ : SUCCESS in 1h 08m 02s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/874690/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/97f82f8/ : SUCCESS in 43m 28s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/874690/2/upstream-check/cinder-tempest-plugin-ontap-nfs/d14a897/ : SUCCESS in 1h 13m 41s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/874690/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/4b251ea/ : FAILURE in 1h 15m 41s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"71acb3abe5dc03dc92ec014b815905580adefb6e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-03-10 14:32:09.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/90/874690/2/check/DellEMC_PowerStore_iSCSI/75b6855/DellEMC_PowerStore_iSCSI/1124 : SUCCESS in 1h 11m 42s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/90/874690/2/check/DellEMC_PowerStore_FC/c7aac06/DellEMC_PowerStore_FC/1246 : SUCCESS in 1h 27m 09s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/90/874690/2/check/DellEMC_PowerStore_NFS/24727de/DellEMC_PowerStore_NFS/763 : SUCCESS in 1h 24m 07s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/90/874690/2/check/DellEMC_PowerStore_NVMeTCP/1f76406/DellEMC_PowerStore_NVMeTCP/627 : SUCCESS in 1h 26m 39s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"f81270327bd53217ecf393a4dc17c0a2961d6a4f","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2023-03-10 14:56:47.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/90/874690/2/check/DellEMC_SC_FC/2c1ae94/DellEMC_SC_FC/1197 : SUCCESS in 1h 08m 02s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/90/874690/2/check/DellEMC_SC_iSCSI/04aaf01/DellEMC_SC_iSCSI/1177 : SUCCESS in 1h 51m 49s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"721464fa8751cdace7da23e51c6d0e1904020000","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-10 15:48:18.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-874690-2/ns5_nfs : FAILURE in 4h 16m 27s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-874690-2/ns5_iscsi : FAILURE in 4h 50m 37s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"47f7c9a2715cd2ccdf7e6832ee04353ae281c8a1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-10 16:10:58.000000000","message":"Patch Set 2:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/75","accounts_in_message":[],"_revision_number":2},{"id":"99b22bf5ef21a5045cff72beb8f53fc639950c80","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-10 19:07:50.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/90/874690/2/check/DellEMC_Unity_FC/7c5bf48/DellEMC_Unity_FC/2474 : SUCCESS in 1h 36m 04s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/90/874690/2/check/DellEMC_Unity_iSCSI/0f6a0b3/DellEMC_Unity_iSCSI/2323 : FAILURE in 2h 38m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"1cf36fa901a3fe1fa600afc4d72f48268157c1f8","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-03-10 22:25:31.000000000","message":"Patch Set 2:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/90/874690/2/fc : FAILURE in 0h 39m 46s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/90/874690/2/iscsi : FAILURE in 0h 39m 10s","accounts_in_message":[],"_revision_number":2},{"id":"d8440f06dad3edad8262a87f16f0b28e2dfc3ea4","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-03-11 23:47:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/874690-2/HPMSA-iSCSI/ : SUCCESS in 1:38:21\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/874690-2/HPMSA-FC/ : SUCCESS in 1:17:47","accounts_in_message":[],"_revision_number":2},{"id":"d32611b70b4d58573fa9f060f097fb243e243c07","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-03-12 19:41:41.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/874690-2/PVME-iSCSI/ : FAILURE in 3:34:40\n- PVME-FC https://os-logs.seagate.com/dell-ci/874690-2/PVME-FC/ : SUCCESS in 1:30:34","accounts_in_message":[],"_revision_number":2},{"id":"ab952acceee14f635f5eac68f544eb1387c9a9b9","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-03-13 06:59:59.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/874690-2/Seagate-iSCSI/ : SUCCESS in 1:40:35\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/874690-2/Seagate-FC/ : SUCCESS in 1:31:00","accounts_in_message":[],"_revision_number":2},{"id":"93e5eb4bc4b084be01eefd0fc4aa5acea3fc434c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-03-16 06:34:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/874690/2/20230316100111/NEC-V-FC : SUCCESS in 2h 45m 17s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/874690/2/20230316100111/NEC-V-ISCSI : SUCCESS in 5h 33m 15s","accounts_in_message":[],"_revision_number":2},{"id":"e7852da34a9ffe17c11b26d01cbe325b93a743d6","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-03-18 17:35:25.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/874690/2/20230318193017/NEC-FC : FAILURE in 7h 01m 45s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/874690/2/20230318193017/NEC-ISCSI : FAILURE in 7h 05m 05s","accounts_in_message":[],"_revision_number":2},{"id":"6c36bb4c041d6a292d7ef8005f333ef2de3c8bd9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-20 06:38:49.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/90/874690/2/check/iscsi-driver-nimble/2bbca8d/2bbca8d.html : FAILURE in 2h 13m 12s","accounts_in_message":[],"_revision_number":2},{"id":"40e926dd1d20a8f8cca09648bb64a318bbc52008","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"date":"2023-07-27 16:31:18.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3304d890f5b5e4bfb6cf276e8fbe761ec67c0fc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 16:31:34.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/os-brick for 874689,4","accounts_in_message":[],"_revision_number":3},{"id":"10957d563767877cfc170de22d5fe52feb620694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 16:31:35.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/os-brick for 874689,4","accounts_in_message":[],"_revision_number":3},{"id":"0d0f26010c0af0feb03042418851152d1be441ec","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-07-27 16:31:49.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"91037ee9164c564735bee585a40db55288a8eb79","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-27 16:31:52.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/os-brick for 874689,4","accounts_in_message":[],"_revision_number":3},{"id":"f0da101b5e164c4693fa2928f89af8f628b3efe3","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-27 16:32:04.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/os-brick for 874689,4","accounts_in_message":[],"_revision_number":3},{"id":"4139ad33b6f8c8134a9a3168d4f6ff90938f8d34","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-07-27 16:32:12.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"cdd0408196a0f2f3fb09d7458bf8e4229da83c15","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-07-27 16:32:13.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"cad98bafc31b5b0dbe2fc18c65922dc7a7347453","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-07-27 17:04:14.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e936579acd62b01a06e5f1455c60d3deac9800c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 17:04:31.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/os-brick for 874689,4","accounts_in_message":[],"_revision_number":3},{"id":"99b16beaf9ffaa8aa4c70adbf12119a9352c4358","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 17:04:31.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/os-brick for 874689,4","accounts_in_message":[],"_revision_number":3},{"id":"5ffcc16a972bcef72172983c3c21a4bf95b54762","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-07-27 17:04:39.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/os-brick for 874689,4","accounts_in_message":[],"_revision_number":3},{"id":"f4fa8327d5a65ea182a725df3bf4b1fea23f5e6e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-07-27 17:04:50.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/os-brick for 874689,4","accounts_in_message":[],"_revision_number":3},{"id":"84a7ff97f5e50e7b6938d3a08350573a55699765","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-27 20:00:12.000000000","message":"Patch Set 3:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/90/874690/3/fc : SUCCESS in 0h 33m 04s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/90/874690/3/iscsi : SUCCESS in 0h 52m 44s","accounts_in_message":[],"_revision_number":3},{"id":"25993d00d13ce23734a48f5175dcdde9fd35df11","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-27 20:22:04.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-874690-3/ns5_nfs : FAILURE in 47m 54s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-874690-3/ns5_iscsi : ABORTED in 2h 48m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"9e68afc611ab4b146c9dc1f90601f1fc88891a81","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-28 05:44:40.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/874690/3/20230728114737/NEC-V-ISCSI : SUCCESS in 2h 36m 52s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/874690/3/20230728114737/NEC-V-FC : SUCCESS in 2h 57m 00s","accounts_in_message":[],"_revision_number":3},{"id":"b649efea81dee9b67cf96841412ec52173042a03","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-28 16:31:08.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/90/874690/3/20230728214828/NEC-FC : SUCCESS in 3h 42m 01s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/90/874690/3/20230728214828/NEC-ISCSI : SUCCESS in 3h 42m 36s","accounts_in_message":[],"_revision_number":3},{"id":"b81a335b00b2b9f30b36cacc568803fbb9ba08cf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-07-31 22:30:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1664/","accounts_in_message":[],"_revision_number":3},{"id":"dcc9cced6317cf7328800f682c397101180813f8","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-08-23 12:35:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"142b3c6c759a0614cd056ed93c3cddbf5a706972","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 12:56:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/101359076c044a6ba4442a69023a4b34\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d9205b014cd4609b553611e1c14151f : SUCCESS in 10m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/060774f56c22456aafcf931e7a5716ab : SUCCESS in 13m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bb20c70fbf404526b69ed464549bfcc7 : SUCCESS in 13m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/10cec7d73e1a4cc583f4126a642df186 : RETRY_LIMIT in 1m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"79d93cb9e56edfb1d2325e32b35238b8936073c2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-08-23 14:33:44.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/dd2e0394fbe643a0b99523ab830f7ad2\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/1f778c5450104aefac5dc20f0520c75d : FAILURE in 1h 19m 48s","accounts_in_message":[],"_revision_number":3},{"id":"4d9e8c866a35884b2c852a09430f1202d1e5f8d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 14:45:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec94e9fcf68c4db297808ed6f7df2f19\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22c1ac1456794d49a979f3eb2ebb15b4 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d93b6458b8a449b19df0b029804b9e8b : SUCCESS in 12m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/553c29ec65ce448182a5aca10055eddb : SUCCESS in 11m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb28e20d3aab4507ae6ea307d57e4237 : SUCCESS in 11m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9cc52df3bf8742ef86c52585165dbb8c : RETRY_LIMIT in 2m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ff4bb7051ad49c6b763822e7f6e4684 : SUCCESS in 10m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/ea4a01d7c0b14c33b52a923fb6a725df : SUCCESS in 1h 02m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2ed7283b7bec477b8e6aa8c0948861b4 : SUCCESS in 1h 27m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/054b69560d2841cbaa69443500a7cdbf : SUCCESS in 48m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12da8dccc49e4749b9ac0ece5dc7260b : SUCCESS in 8m 16s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/816b384c68f94b7d9e9b911d1baa5986 : SUCCESS in 16m 25s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/57719c666c3643fd9204b8a54a791535 : SUCCESS in 5m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/be4a4b0a2bfb4590b4644982f1b027fc : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a29e560a8547463096fd320c5ac1c70e : SUCCESS in 6m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/87d1bdb2576649ac9b32b76fabb2e32f : SUCCESS in 6m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/709864e8537c45fea0dc40fca7c7a913 : SUCCESS in 54m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8303ee9deaef44feb4b663307bdd1bbc : SUCCESS in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a22f90d8ad7b4524b92e163fe0e7cb74 : SUCCESS in 1h 54m 42s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/13263b9e27984523a59f97618114e8c9 : FAILURE in 2h 01m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cc1dba592ef54eada4e9d627ff77df35 : SUCCESS in 1h 47m 44s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/6cbb13234e3a4e43a5409b52b1c81619 : FAILURE in 40m 15s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9080521a0b1c4a8796af2e8279b72add : SUCCESS in 1h 16m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/d47249b3689e4984b26ee23849965f0e : SUCCESS in 1h 25m 15s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c746061dcadb4daba7d379dc2a5978fc : SUCCESS in 43m 50s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d40e70641476415f810b8fb72d1f0c6f : SUCCESS in 1h 24m 12s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4d6527da910a4ee4b0ac41a960033e7f : FAILURE in 36m 33s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8e6adc93a16c4ad2a098a6b8ddb86653 : SUCCESS in 1h 13m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7696e43bb1244bd7b091af8e389818e2 : SUCCESS in 57m 39s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/0ac6ee98aeba4c058178dc1bceb61450 : SUCCESS in 11m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"431dd85c435f6d9d4ac69c500c1fb7f2a7b5a3b4","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-08-23 14:47:20.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"a4c0a5f6bced03a155acdfe2d840bb65dd18853b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-23 16:16:26.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/874690/3/upstream-check/cinder-tempest-plugin-ontap-fcp/6415008/ : FAILURE in 1h 29m 00s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/874690/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/ac1f22c/ : SUCCESS in 1h 19m 11s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/874690/3/upstream-check/cinder-tempest-plugin-ontap-nfs/5da2971/ : SUCCESS in 45m 18s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/874690/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/bd1f257/ : SUCCESS in 1h 19m 49s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"92bae534e0a2f3e4faee16b1442b1afed253e66f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-23 17:48:30.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/90/874690/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/22b773e/ : SUCCESS in 2h 08m 53s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/90/874690/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/4628d12/ : SUCCESS in 1h 54m 17s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/90/874690/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/f1fadb4/ : SUCCESS in 1h 52m 56s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/90/874690/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/331dbc5/ : SUCCESS in 2h 31m 43s","accounts_in_message":[],"_revision_number":3},{"id":"f1bc04d23e75d28c9ce2a5e0bb228cefe664bb20","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-23 23:30:19.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/2909b9e41cf04d048db5106c243a0a1d : SUCCESS in 1h 12m 11s\n- cinder-storpool-tempest-iscsi-multipath https://spfactory.storpool.com/zuul/t/local/build/a4ed40bbe446498f938e64615d2ff89b : RETRY_LIMIT in 57m 02s","accounts_in_message":[],"_revision_number":3},{"id":"e2d0465963ce303acec44862d4864ab7c728c6bc","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-08-24 10:30:24.000000000","message":"Patch Set 3:\n\nBuild failed.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/874690-3/HPMSA-iSCSI/ : FAILURE in 1:20:26\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/874690-3/HPMSA-FC/ : FAILURE in 1:08:16","accounts_in_message":[],"_revision_number":3},{"id":"2e0167c6c1414100508620a97d440bdef540bec4","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-09-14 14:17:55.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"22b92ba62e9be8a1db6c4c0bb61d79cba920f216","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-09-15 14:45:14.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9f10cc701897c7576637130e13b07591f40bdc83","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-09-15 14:52:11.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4478a88d4c02f2296f10e7870be4ca6a8b52b34d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-09-15 17:20:18.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"69c8f18b4432ae0f1d1cdf37c28f8d3f3a20b362","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-09-15 17:21:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e4664a45c2446a3f08a115ad120932d4e27a2562","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-15 17:30:06.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8fa14d660a8241a6b267fa17ff58fbeb : FAILURE in 7m 50s","accounts_in_message":[],"_revision_number":4},{"id":"31b60597de92f83a93987e90e9e9a6520a6adcc3","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-09-15 18:41:52.000000000","message":"Patch Set 4:\n\nBuild failed (iSCSI). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1851/","accounts_in_message":[],"_revision_number":4},{"id":"8994009da4319e3d4cc7ea2c121b057e345936ce","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-09-15 18:46:56.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/2bada28de32b4e86a5eabf95dab213ef\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/f2a257bb934e4e46a3fb8474b5c540d5 : FAILURE in 1h 23m 44s","accounts_in_message":[],"_revision_number":4},{"id":"fbb2ae10128ddc380876e31922b413538bd45dad","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-09-15 19:00:06.000000000","message":"Patch Set 4:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/90/874690/4/fc : SUCCESS in 0h 26m 17s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/90/874690/4/iscsi : SUCCESS in 0h 39m 51s","accounts_in_message":[],"_revision_number":4},{"id":"dc6ea50dc52443683ddde19306e28c8ab150e08d","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-09-15 19:04:30.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/90/874690/4/check/fujitsu-eternus-dx-iscsi/cc4a829 : FAILURE in 51m 53s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/90/874690/4/check/fujitsu-eternus-dx-fc/59d52d4 : SUCCESS in 1h 43m 30s","accounts_in_message":[],"_revision_number":4},{"id":"2cbf998594a235b78d799055d007a085447f7b53","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-15 19:14:04.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/90/874690/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/afd13d1/ : FAILURE in 1h 45m 04s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/90/874690/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/399b2fb/ : SUCCESS in 1h 41m 05s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/90/874690/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/a3f549e/ : SUCCESS in 1h 52m 00s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/90/874690/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/8756f54/ : SUCCESS in 1h 36m 32s","accounts_in_message":[],"_revision_number":4},{"id":"d6e72463e667deb08ae0c8c5bba95adb73f5aeec","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-15 19:15:14.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/90/874690/4/check/DellEMC_Unity_FC/6ac1602/DellEMC_Unity_FC/801 : SUCCESS in 1h 54m 34s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/90/874690/4/check/DellEMC_Unity_iSCSI/f617c55/DellEMC_Unity_iSCSI/811 : FAILURE in 22m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"2882107cb5196d2efa61048bf49cfae2923cbe0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 19:26:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7972518fb5d48ed84818fd223e80e75\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8bd7990f1f0547389f72b2906f7bff5c : SUCCESS in 27m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b0be508868044f7a6ff80fe510018b7 : SUCCESS in 21m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/715ddb39b0bb409dbc79e6e5eac9390d : SUCCESS in 28m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/062696e5344d46358c061b8e3ab4a6f2 : SUCCESS in 26m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"df1fdfe9e415171401a31f6923efbe8ec8d51345","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-09-15 19:45:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (FC). \nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/638/","accounts_in_message":[],"_revision_number":4},{"id":"77cef4836db8e0dd626aea2388c3fededa5e048d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 19:48:23.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/05f92810da934a3dacbddc3b10a1f71e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0377d28cd20346ce965c198ebfb8b743 : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b73789505def44729519b5c084df8030 : SUCCESS in 10m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22f0bcd8bb3d4466817fba47809a933b : SUCCESS in 12m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9631957a17d34483bbf33c2dbde360ad : SUCCESS in 12m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17bb802aa2b24b828901287a8f4c7e56 : SUCCESS in 13m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc99203f3b6a4e43b8ad07aa99acd4d8 : SUCCESS in 15m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/8efb830ea7094832a3eb0cdedb643570 : SUCCESS in 1h 19m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4e3ce89e012743bcacf82260bc91d154 : SUCCESS in 1h 32m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2aecfe041d4e42e1b82c80260432d5dc : SUCCESS in 1h 18m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49d57974534d46b18d20a9750087b6dc : SUCCESS in 8m 50s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/3f082dc07bf840139b1cff3d194b7dd6 : SUCCESS in 16m 37s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/7d147fae86134929a83b8a38fbeae360 : SUCCESS in 5m 41s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/69417d59a90c4e7ebd633e0f4ed3cacb : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/08374f43d5704b0cb7ec2199c8a97033 : SUCCESS in 7m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9dd7defd9a43408ba16bef9c6e9dd6ff : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5440ad626f204df3b58b2b1b9af5fa28 : SUCCESS in 1h 04m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8ff329d4e8cc435cb4181fb52788669e : SUCCESS in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/19dd751df6cf412eb59482b7f6d6c011 : SUCCESS in 2h 18m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/01ea83962e3942faac5026cb7e7d139b : SUCCESS in 1h 56m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7afdeaf4e1d849cd825a25f703ee7268 : POST_FAILURE in 1h 26m 41s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/36751e065b554bd69e003439a5a819a4 : SUCCESS in 2h 08m 59s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9ae0e9382b70468f9cce838216e9f719 : SUCCESS in 1h 06m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/33d5c5f14e4148c3958be8647196dcc3 : FAILURE in 1h 03m 31s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9248038a6d3a41e0b650bd743919d1e8 : SUCCESS in 52m 14s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/3400f0f1b97d4df9ae75a900e921d3e4 : SUCCESS in 1h 26m 04s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/968b9794e2e14dbb8df4e838e1901d17 : POST_FAILURE in 2h 21m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7f9af77cc8df4f988ace0a39697743ad : SUCCESS in 1h 12m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e559d8399edf4313bf8ed5d96dfe55e7 : SUCCESS in 52m 46s","accounts_in_message":[],"_revision_number":4},{"id":"612ef8b77d2532fad18a215f48f4786d7c8915bb","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-09-15 19:54:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"df32d16eb2743b66abdc0afd5a146b65d187b6f8","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-15 20:04:00.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/598059a0dcf34c88b3009c83716f3683 : FAILURE in 8m 07s","accounts_in_message":[],"_revision_number":4},{"id":"8ca2388bd50e56cff91e503c8e266e0ed0159ebf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 20:23:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea7970608a4a48b0a2f5326724ff2e38\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c18322901efc4eeb93af185f8d4a4599 : SUCCESS in 24m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d051249d5c34468a130514551b93427 : SUCCESS in 21m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4306d41d3c07479b823afb2987baa1b3 : SUCCESS in 24m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/14a0ec7d635a4dbcbd3c4c147b5326ae : SUCCESS in 22m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b114e82ed815f6380a24de97123f581b19ec2c9f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-09-15 20:26:13.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-90-874690-4/FC/ : SUCCESS in 2h 28m 39s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-90-874690-4/iSCSI/ : FAILURE in 3h 5m 35s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"430a381443930f101ecf764296e12a7f4fc1ae10","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-09-15 21:13:03.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/874690 : FAILURE in 25m 23s","accounts_in_message":[],"_revision_number":4},{"id":"1015ecb7c882b2c2d97ea4bafe910920c1379a50","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-09-15 21:17:38.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-90-874690-4/ns5_nfs : SUCCESS in 3h 52m 14s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-90-874690-4/ns5_iscsi : SUCCESS in 3h 56m 49s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"d629a4d92307f74084ddf6f693a4f4cb0d45298c","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-09-15 21:26:17.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/fefb1fde94d0423ba447219e0d66c3f4\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/ce1d396f233c438bb5952c0496b7129f : FAILURE in 1h 22m 11s","accounts_in_message":[],"_revision_number":4},{"id":"0131e82e59f9a3df18958ef29bd8465fe47d6ea8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-15 21:40:19.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/90/874690/4/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/7cb74b6/ : SUCCESS in 1h 42m 34s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/90/874690/4/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/68d2bcd/ : SUCCESS in 1h 44m 52s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/90/874690/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/291d2dd/ : SUCCESS in 1h 40m 28s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/90/874690/4/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/5878834/ : SUCCESS in 1h 39m 47s","accounts_in_message":[],"_revision_number":4},{"id":"b4c0b0867d68abf0367a42d342bf0dde272f1f9b","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-09-15 21:47:00.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/90/874690/4/check/fujitsu-eternus-dx-iscsi/838a823 : FAILURE in 54m 35s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/90/874690/4/check/fujitsu-eternus-dx-fc/1b71001 : SUCCESS in 1h 52m 13s","accounts_in_message":[],"_revision_number":4},{"id":"14951a5497fb25e460e72f811aac25bc87f445c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 22:36:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12d3f854d9774d5cb600cb8a2dceef4f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34b2697ae065448fa96c388600674823 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e09d2007fb6e4c0f972d5ed600c118b3 : SUCCESS in 12m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47fc2568a29a4820ae7f31d279f0dd78 : SUCCESS in 8m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bef19a20d35443129bb76ba9955d7f26 : SUCCESS in 11m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/383ad0b4511e4356b2e874ed5d6b4ec3 : SUCCESS in 12m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d6b68d85d3a4ff9bf2fb41bf9f3e631 : SUCCESS in 12m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/ef838c4388f04b7590f4abfc3b58f6d6 : SUCCESS in 47m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1fd69c5662084dbf9569d59070986e09 : SUCCESS in 1h 00m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aaf655bed8f54dd58d448a061beeb3e1 : SUCCESS in 1h 03m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e508661d32fa43a1a6fedc1ce81c4765 : SUCCESS in 8m 08s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e99d70b995ca4a619e41dbd6cfea8ef2 : SUCCESS in 13m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e7ec866793644907bcc9fdc4c9293b2a : SUCCESS in 4m 28s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0801c733914f41218003aaddf9900842 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8114647015144912bcfdb474a016caef : SUCCESS in 5m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4f5322449ee04889959909f9a6ca6cee : SUCCESS in 7m 18s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/154f759d118843fd93fa480773ca26a2 : SUCCESS in 42m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e9f06afff6a9466dadd639fd7cb49ac9 : SUCCESS in 6m 14s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f645d310cdf3409a99b4d46ca162073b : SUCCESS in 2h 10m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/b772514a0c8e43ccbd906dc3e9eac767 : SUCCESS in 1h 38m 32s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fa94c3c658a84d179397299d1f7fd126 : SUCCESS in 1h 49m 24s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/183d519d870f472e88d1631388fa84ff : SUCCESS in 1h 54m 21s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/4a3bef9a2b4c4bb2a4680d65571dfef4 : SUCCESS in 1h 10m 26s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e06fb24f384e457b9df1662d26b7b2af : SUCCESS in 1h 23m 14s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f28f88550d1e43daad6cc6300b8a4599 : SUCCESS in 46m 33s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2053fa38fcf24b159603f088bf5092b3 : SUCCESS in 1h 23m 42s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/ae33f8394e4c4a5cb8783efe08366c80 : POST_FAILURE in 2h 35m 29s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2665a6613811461f9687ecdce33ed17b : SUCCESS in 1h 40m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1b28d0b7c8846199eb912bcbffeac6c : SUCCESS in 39m 10s","accounts_in_message":[],"_revision_number":4},{"id":"36345f6bc603fe0be9636f2ece6a43ced7011d2c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-09-15 23:32:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"08e3d25ab77a963a940ecd556939dffc42a2e44a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 23:33:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1961bd1004fa1b1f42156d27e392dcfa99e9b016","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-16 00:26:08.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/874690/4/upstream-check/cinder-tempest-plugin-ontap-fcp/bf77314/ : SUCCESS in 1h 04m 07s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/874690/4/upstream-check/cinder-tempest-plugin-ontap-iscsi/6647ba1/ : SUCCESS in 1h 43m 33s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/874690/4/upstream-check/cinder-tempest-plugin-ontap-nfs/379a2f7/ : SUCCESS in 1h 47m 48s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/90/874690/4/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/89d9302/ : SUCCESS in 1h 42m 52s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"5250859d325ba4fd0ef9977f07a1f03abcd60dc9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-16 00:40:41.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/90/874690/4/check/DellEMC_Unity_FC/6729b09/DellEMC_Unity_FC/802 : SUCCESS in 1h 51m 06s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/90/874690/4/check/DellEMC_Unity_iSCSI/10c2fc7/DellEMC_Unity_iSCSI/812 : SUCCESS in 2h 03m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"bed264cf68609c03321d8939e5cc5d2f79f339e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-16 02:26:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6025d3fb385a47a2baa90faa6071e352\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a872a9ea4d11412e9d00b766cbe42a06 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44f3cf2749e34d70b31bc4a247d94829 : SUCCESS in 9m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da6ae98e7ae54cb9b1423d1630b057e3 : SUCCESS in 9m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eacd0fb4df14495ea17759691be6cbca : SUCCESS in 10m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ab1ccc4f7c44cf8bc0ca0cc8a1b0d9e : SUCCESS in 10m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/d858d9fe713a45c0849a95d472ab1e59 : SUCCESS in 40m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/16f1a50572d0436e83dc75fb204ef493 : SUCCESS in 55m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/baeffaeba4674411b584ae495df7eb0e : SUCCESS in 41m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b4d4034273244ffa6de2a334b826574 : SUCCESS in 7m 06s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/0e05010d068b4745a907c5f1ca9a655d : SUCCESS in 1h 14m 09s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cfae5360f31e432c808fda517407dc24 : SUCCESS in 1h 57m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b084f63b069443b9bacb1794f319d410 : SUCCESS in 36m 40s","accounts_in_message":[],"_revision_number":4},{"id":"afcf92613df57d27fdaa423e64bdb6a008667517","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-16 02:26:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"43f0a5a7ac3b878e943c51f7b16af72d7e679aa1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-16 02:29:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29f43d4fb5fd48babaec44651abb03ca\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8517026c4f7b4294bd1bf37ac74ecad7 : SUCCESS in 1m 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9509e2bebd904f91913c2a67c603fbcf : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":4},{"id":"d4cd283d42d634ee3d672ec9977aa1c864160106","tag":"autogenerated:lenovo-ci","author":{"_account_id":24815,"name":"Lenovo Storage CI","email":"cinder+lenovo@tristero.net","username":"lenovo-ci"},"date":"2023-09-17 13:09:22.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- Lenovo-iSCSI https://os-logs.seagate.com/lenovo-ci/874690-4/Lenovo-iSCSI/ : SUCCESS in 2:07:12\n- Lenovo-FC https://os-logs.seagate.com/lenovo-ci/874690-4/Lenovo-FC/ : SUCCESS in 1:41:47","accounts_in_message":[],"_revision_number":4},{"id":"6b17a7c65ea1a16c224c47a95b01947e6a1b72de","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-09-18 15:36:54.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/874690-4/HPMSA-iSCSI/ : SUCCESS in 1:39:29\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/874690-4/HPMSA-FC/ : SUCCESS in 1:23:39","accounts_in_message":[],"_revision_number":4},{"id":"41880ee9d4949b4368353efc9d68448127056da5","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-09-19 11:10:57.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/874690-4/PVME-iSCSI/ : SUCCESS in 2:04:29\n- PVME-FC https://os-logs.seagate.com/dell-ci/874690-4/PVME-FC/ : SUCCESS in 1:41:58","accounts_in_message":[],"_revision_number":4},{"id":"907ed50f18ffb66fb584a88fa88298d04137b8ca","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-20 04:22:32.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/90/874690/4/check/DellEMC_PowerFlex/7b682ac/DellEMC_PowerFlex/622 : FAILURE in 1s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"97c182d941dc02394c5f71ee485a47c49d1867eb","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-20 16:19:30.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/90/874690/4/check/DellEMC_PowerStore_iSCSI/bd37fe4/DellEMC_PowerStore_iSCSI/304 : FAILURE in 20s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/90/874690/4/check/DellEMC_PowerStore_FC/d1c9bf7/DellEMC_PowerStore_FC/313 : SUCCESS in 1h 12m 33s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/90/874690/4/check/DellEMC_PowerStore_NFS/eaf4912/DellEMC_PowerStore_NFS/312 : SUCCESS in 44m 53s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/90/874690/4/check/DellEMC_PowerStore_NVMeTCP/20a41ae/DellEMC_PowerStore_NVMeTCP/301 : FAILURE in 20s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"dad485ea665271be66beca750025e3def2a9dc05","revisions":{"02be1aa0be65294dbc1097a31b4b9706a3ef64e6":{"kind":"REWORK","_number":1,"created":"2023-02-21 19:23:57.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/90/874690/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/874690/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/874690/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/874690/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/874690/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/874690/1"}}},"commit":{"parents":[{"commit":"a92aa06e463f6322e8abdc2f570a801ea82dd1cb","subject":"Hitachi: add GAD volume support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a92aa06e463f6322e8abdc2f570a801ea82dd1cb"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2023-02-21 19:17:32.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2023-02-21 19:18:14.000000000","tz":60},"subject":"Pure: Report SAM-2 addressing mode for LUNs","message":"Pure: Report SAM-2 addressing mode for LUNs\n\nPure iSCSI and FC drivers use SCSI SAM-2 addressing mode, which means\nthat LUNs \u003c 256 use peripheral mode (unmodified values) and LUNs \u003e\u003d 256\nuse flat mode (2 higher bits of the MSB must be 01b).\n\nThis is not the standard behavior of os-brick, which defaults to\nSAM/transparent mode.\n\nIn this patch the Pure storage driver reports it\u0027s addressing mode so\nthat LUNs with value greater than 255 can be connected by os-brick.\n\nDepends-On: If32d054e8f944f162bdc8700d842134a80049877\nCloses-Bug: #2006960\nChange-Id: I4abfb6de58340c5ea70e7a796326f5d2089c80eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/02be1aa0be65294dbc1097a31b4b9706a3ef64e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/02be1aa0be65294dbc1097a31b4b9706a3ef64e6"}]},"branch":"refs/heads/master"},"cf85aa40b74eefad58288d3e1dedf9d22c605311":{"kind":"REWORK","_number":2,"created":"2023-03-10 10:52:52.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/90/874690/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/874690/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/874690/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/874690/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/874690/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/874690/2"}}},"commit":{"parents":[{"commit":"a92aa06e463f6322e8abdc2f570a801ea82dd1cb","subject":"Hitachi: add GAD volume support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/a92aa06e463f6322e8abdc2f570a801ea82dd1cb"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2023-02-21 19:17:32.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2023-03-10 10:52:41.000000000","tz":60},"subject":"Pure: Report SAM-2 addressing mode for LUNs","message":"Pure: Report SAM-2 addressing mode for LUNs\n\nPure iSCSI and FC drivers use SCSI SAM-2 addressing mode, which means\nthat LUNs \u003c 256 use peripheral mode (unmodified values) and LUNs \u003e\u003d 256\nuse flat mode (2 higher bits of the MSB must be 01b).\n\nThis is not the standard behavior of os-brick, which defaults to\nSAM/transparent mode.\n\nIn this patch the Pure storage driver reports it\u0027s addressing mode so\nthat LUNs with value greater than 255 can be connected by os-brick.\n\nDepends-On: If32d054e8f944f162bdc8700d842134a80049877\nCloses-Bug: #2006960\nChange-Id: I4abfb6de58340c5ea70e7a796326f5d2089c80eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf85aa40b74eefad58288d3e1dedf9d22c605311"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cf85aa40b74eefad58288d3e1dedf9d22c605311"}]},"branch":"refs/heads/master"},"ca587487e4f54b26469ce811fe06813b8bb9a7ac":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-07-27 16:31:18.000000000","uploader":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"ref":"refs/changes/90/874690/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/874690/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/874690/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/874690/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/874690/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/874690/3"}}},"commit":{"parents":[{"commit":"4da12c70ac06cb66159d433e42856e4ac92d39b3","subject":"Merge \"Automate generation of backups api-ref samples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4da12c70ac06cb66159d433e42856e4ac92d39b3"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2023-02-21 19:17:32.000000000","tz":60},"committer":{"name":"Matthew Heler","email":"matthew.heler@hotmail.com","date":"2023-07-27 16:31:18.000000000","tz":0},"subject":"Pure: Report SAM-2 addressing mode for LUNs","message":"Pure: Report SAM-2 addressing mode for LUNs\n\nPure iSCSI and FC drivers use SCSI SAM-2 addressing mode, which means\nthat LUNs \u003c 256 use peripheral mode (unmodified values) and LUNs \u003e\u003d 256\nuse flat mode (2 higher bits of the MSB must be 01b).\n\nThis is not the standard behavior of os-brick, which defaults to\nSAM/transparent mode.\n\nIn this patch the Pure storage driver reports it\u0027s addressing mode so\nthat LUNs with value greater than 255 can be connected by os-brick.\n\nDepends-On: If32d054e8f944f162bdc8700d842134a80049877\nCloses-Bug: #2006960\nChange-Id: I4abfb6de58340c5ea70e7a796326f5d2089c80eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ca587487e4f54b26469ce811fe06813b8bb9a7ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/ca587487e4f54b26469ce811fe06813b8bb9a7ac"}]},"branch":"refs/heads/master","description":"Rebase"},"dad485ea665271be66beca750025e3def2a9dc05":{"kind":"REWORK","_number":4,"created":"2023-09-15 17:20:18.000000000","uploader":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"ref":"refs/changes/90/874690/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/90/874690/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/874690/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/874690/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/90/874690/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/90/874690/4"}}},"commit":{"parents":[{"commit":"4da12c70ac06cb66159d433e42856e4ac92d39b3","subject":"Merge \"Automate generation of backups api-ref samples\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4da12c70ac06cb66159d433e42856e4ac92d39b3"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2023-02-21 19:17:32.000000000","tz":60},"committer":{"name":"Simon Dodsley","email":"simondodsley@gmail.com","date":"2023-09-15 17:20:07.000000000","tz":-240},"subject":"Pure: Report SAM-2 addressing mode for LUNs","message":"Pure: Report SAM-2 addressing mode for LUNs\n\nPure iSCSI and FC drivers use SCSI SAM-2 addressing mode, which means\nthat LUNs \u003c 256 use peripheral mode (unmodified values) and LUNs \u003e\u003d 256\nuse flat mode (2 higher bits of the MSB must be 01b).\n\nThis is not the standard behavior of os-brick, which defaults to\nSAM/transparent mode.\n\nIn this patch the Pure storage driver reports it\u0027s addressing mode so\nthat LUNs with value greater than 255 can be connected by os-brick.\n\nDepends-On: If32d054e8f944f162bdc8700d842134a80049877\nCloses-Bug: #2006960\nChange-Id: I4abfb6de58340c5ea70e7a796326f5d2089c80eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dad485ea665271be66beca750025e3def2a9dc05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dad485ea665271be66beca750025e3def2a9dc05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
