)]}'
{"id":"openstack%2Fcinder~877717","triplet_id":"openstack%2Fcinder~master~If1ef9bceae96ef044209c758eee2fd22b44efa49","project":"openstack/cinder","branch":"master","topic":"powermax_check_force","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2023-03-17 23:31:54.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-03-17 23:31:54.000000000","reason":"Change was submitted"},"12670":{"account":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"last_update":"2023-03-17 23:31:54.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-03-17 19:34:50.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"If1ef9bceae96ef044209c758eee2fd22b44efa49","subject":"Make PowerMax tests stable","status":"MERGED","created":"2023-03-16 20:40:18.000000000","updated":"2023-04-19 23:40:16.000000000","submitted":"2023-03-17 23:31:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"877717-powermax_check_force","meta_rev_id":"18603d74a56bc4ac8f3d2ba1e5c634f5a28e500a","_number":877717,"virtual_id_number":877717,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-17 23:31:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2023-03-17 20:05:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-17 19:34:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2023-03-17 20:05:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-16 20:40:25.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-03-16 20:41:31.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":12670,"name":"Helen Walsh","email":"helen.walsh@emc.com","username":"walshh2"},"state":"REVIEWER"},{"updated":"2023-03-16 20:49:57.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-16 21:11:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-16 21:15:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-16 21:23:14.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-03-16 21:58:43.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-16 22:39:50.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-03-16 22:46:57.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2023-03-16 22:48:29.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-03-16 22:49:18.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-16 23:07:26.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-03-17 00:08:04.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-17 01:00:04.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-03-17 01:13:53.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-03-17 03:27:22.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-03-17 03:28:55.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2023-03-17 13:41:59.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2023-03-17 15:06:32.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-03-17 15:13:38.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"98091c671e1d358aac5922f1bfa8f98d6d40eada","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-03-16 20:40:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d89f0962ebb645e29d2030e43a82561062c3474a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-16 20:40:25.000000000","message":"Patch Set 1:\n\nBuild Started http://openstack-ci.spb.yadro.com:8080/job/openstack-gerrit-trigger-fc/130/ (1/2)","accounts_in_message":[],"_revision_number":1},{"id":"bf755a15a2cb3c7573db347f29ea904e0dbd447c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-16 20:47:05.000000000","message":"Patch Set 1:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/1193","accounts_in_message":[],"_revision_number":1},{"id":"1b3f0dbecdac8ddddfa37a8f9c305e11ced8f746","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-16 20:49:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder iscsi-driver-primera-cinder : RETRY_LIMIT in 0s","accounts_in_message":[],"_revision_number":1},{"id":"6e6e503ea310f084dd1a51cd5f6b7534e3811a94","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-16 21:09:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/877717 : FAILURE in 27m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e42d8a719a88c6cddadf32bb97a2ecd1c809657c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 21:11:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fab3e4ccee8a4739b89cb3c5cb83c3f1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/542bf2606af54bed9dbb351e0cec6c56 : SUCCESS in 10m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/80abc061111e4a5f9ebc38b0d8719aad : SUCCESS in 28m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c81652b75c5ad51007711f0f5ca007f3c9be223","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 21:15:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4791a928bda947dfa25b4c7a4812c1d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8152a0f3a57e4ad99a864e08b4cc1811 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2aa42fe5464d45dd883fc9280f34770a : SUCCESS in 11m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/813692c78313417ca430cb83caf846fe : SUCCESS in 11m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ef3ab95fa3047d5a33284c3a5bf3231 : SUCCESS in 12m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f649cc510511481eaa6275a3305f16cc : SUCCESS in 22m 28s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/03b8833a0d934362a086b4e48dada9ee : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"9ccb4a497d0c91bc0f173f273999d27fda3803ee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-03-16 21:23:14.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/16_03_2023_21_15_19\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"8384d289cf516c5decac2236b87e8d3b6b921392","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-16 21:58:43.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/796e7bc5fddb4e04b425ff804b321442 : SUCCESS in 1h 17m 04s","accounts_in_message":[],"_revision_number":1},{"id":"4ba4e5bc5b9e8e7de239e9b8f55fcb91e1b2490d","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-16 22:39:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/17/877717/1/check/iscsi-driver-alletra-6k/5924173/5924173.html : FAILURE in 1h 58m 26s","accounts_in_message":[],"_revision_number":1},{"id":"2d39bf967c337f81e89e9a69d82a6cdcb4186248","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2023-03-16 22:46:57.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/17/877717/1/check/DellEMC_SC_FC/0a836c3/DellEMC_SC_FC/1219 : SUCCESS in 1h 12m 35s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/17/877717/1/check/DellEMC_SC_iSCSI/d03c799/DellEMC_SC_iSCSI/1199 : SUCCESS in 1h 31m 28s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":1},{"id":"dab320d5116192016cefc5f0c0e978fc2e0c63ae","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-03-16 22:48:29.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/17/877717/1/check/DellEMC_PowerStore_iSCSI/6147ac2/DellEMC_PowerStore_iSCSI/1146 : SUCCESS in 1h 17m 42s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/17/877717/1/check/DellEMC_PowerStore_FC/52f3d06/DellEMC_PowerStore_FC/1268 : SUCCESS in 1h 32m 58s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/17/877717/1/check/DellEMC_PowerStore_NFS/7b2ddf3/DellEMC_PowerStore_NFS/785 : SUCCESS in 1h 20m 44s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/17/877717/1/check/DellEMC_PowerStore_NVMeTCP/45b66a1/DellEMC_PowerStore_NVMeTCP/649 : SUCCESS in 1h 25m 45s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"2b84cf30726f72e2760e86fe925abb6cbae4b074","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-03-16 22:49:18.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/17/877717/1/check/DellEMC_PowerFlex/e5a6f08/DellEMC_PowerFlex/1753 : SUCCESS in 1h 33m 49s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"ac2ec9467911e50325238c33826e3563461ce618","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-16 23:07:26.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/9c9e0938edac455fa5511c891f43f9a2 : SUCCESS in 1h 11m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b01d579fb9563f0281c0ecedc919a035f86582ff","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-17 00:08:04.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-17-877717-1/ns5_nfs : SUCCESS in 3h 23m 45s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-17-877717-1/ns5_iscsi : SUCCESS in 3h 27m 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"8900116d40ba527a6fd270743172550880f5e923","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-03-17 01:00:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/877717-1/HPMSA-iSCSI/ : SUCCESS in 1:36:02\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/877717-1/HPMSA-FC/ : SUCCESS in 1:17:54","accounts_in_message":[],"_revision_number":1},{"id":"674ab5af4f8e3ffcf30e4cf5d1dfad6c41504e29","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-17 01:13:53.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/17/877717/1/check/DellEMC_Unity_FC/3b63d09/DellEMC_Unity_FC/2515 : SUCCESS in 1h 48m 26s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/17/877717/1/check/DellEMC_Unity_iSCSI/d1f2b62/DellEMC_Unity_iSCSI/2363 : SUCCESS in 1h 55m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"e3a23616d256d0042aae05f5116ef05b6952c7f5","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-03-17 03:27:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/877717-1/PVME-iSCSI/ : FAILURE in 1:40:58\n- PVME-FC https://os-logs.seagate.com/dell-ci/877717-1/PVME-FC/ : SUCCESS in 1:33:23","accounts_in_message":[],"_revision_number":1},{"id":"bef87c8b8a745f211d569bace2a3eda86abfd0c3","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-03-17 03:28:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/877717-1/Seagate-iSCSI/ : SUCCESS in 1:47:10\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/877717-1/Seagate-FC/ : SUCCESS in 1:33:44","accounts_in_message":[],"_revision_number":1},{"id":"0061a75ad97af9ae4e040ef7e5eb55b3f8b3a516","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-03-17 13:41:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a1fe874e0050a1b6d9b7d9bf28bf6279f7931468","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-03-17 15:06:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"bdb3f1d5a2130a9e66ed81ef71a952394eeb8883","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-03-17 15:13:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5fe840c62de491f4164d69661929817c6376d48d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-03-17 15:13:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"293244310f2e9e5ce2b91d1f61b61b3e4f145414","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 15:14:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5adb899168fd70be2590fac554cbf3fc64b4ecf9","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-03-17 15:14:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"546a4a166d3396d89926fbb95991927def48844a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-03-17 15:16:52.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"53d01718516af0d91b490068be1a46d82852921a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-17 15:16:59.000000000","message":"Patch Set 2:\n\nBuild Started http://openstack-ci.spb.yadro.com:8080/job/openstack-gerrit-trigger-fc/138/ (2/2)","accounts_in_message":[],"_revision_number":2},{"id":"3db4177de23a4a5446d01af9cf7b21640a9db530","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-03-17 15:19:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3646cf32bbdd18f83a46b89bba4e69aa04a9581d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-17 15:23:42.000000000","message":"Patch Set 2:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/1200","accounts_in_message":[],"_revision_number":2},{"id":"c0a3eb37fb97642dbe87d6013ec70dfe117a6389","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-17 15:24:19.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder finger://openstack-ci-fc/21ae49f4eba34d15a7ad779a2481c8fe : RETRY_LIMIT in 1m 19s","accounts_in_message":[],"_revision_number":2},{"id":"5fe4309dd336e884103ba2195e1919761b5a9709","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 15:30:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf05119bee3b4226801f87b49364b7e4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8b6fc33b17e8400aa517be71efe71c37 : SUCCESS in 12m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/88c0c116cca7440d850840063056894b : SUCCESS in 11m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"844b50256a6dae9c6e4542a40e477e8d2edb2b0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 15:34:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4d0e2d07f5849f6b117f69d563a4f45\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fea21c083f54d60a048bbdc875d22b7 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d8dd9fb83ec483a88d707e2efda8a1a : SUCCESS in 14m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78b7ce4c226d46b49e69a439621d2385 : SUCCESS in 11m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfa7ed9b003445aea3dabdd651a10f48 : SUCCESS in 14m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/90a1a2bb6e5f4086bdd0add231e69b72 : SUCCESS in 15m 54s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/947fb1449d364acb828ba38d3b7edf0c : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":2},{"id":"8f2c9f45b153101801c87b158d56095066babafa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-03-17 15:42:39.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_03_2023_15_34_46\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"056ab0708f947e92a13e1196163f4bd412fa4fa7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-17 15:47:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/877717 : FAILURE in 27m 42s","accounts_in_message":[],"_revision_number":2},{"id":"c6b9223cb459ceb8ea3277345692cf9166236dd5","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-17 16:33:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/85ae0494109345b2ab30425eb29bfdd0 : SUCCESS in 1h 15m 39s","accounts_in_message":[],"_revision_number":2},{"id":"81aca08ebd07d04c025c715c49c842c6badcd826","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-17 16:33:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/6a74685735df447485d914e7d59d76b1 : SUCCESS in 1h 14m 45s","accounts_in_message":[],"_revision_number":2},{"id":"e445852cec867bee6793db5fad0d7e8b33821181","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-03-17 17:08:15.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/17/877717/2/check/DellEMC_PowerStore_iSCSI/d27bdf5/DellEMC_PowerStore_iSCSI/1149 : SUCCESS in 1h 21m 47s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/17/877717/2/check/DellEMC_PowerStore_FC/434b91b/DellEMC_PowerStore_FC/1271 : SUCCESS in 1h 33m 17s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/17/877717/2/check/DellEMC_PowerStore_NFS/e75ecda/DellEMC_PowerStore_NFS/788 : SUCCESS in 1h 16m 53s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/17/877717/2/check/DellEMC_PowerStore_NVMeTCP/a419d32/DellEMC_PowerStore_NVMeTCP/652 : SUCCESS in 1h 32m 13s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"5b88364e8a9eb26e00e3a23340a83430fb2573e7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-03-17 17:20:33.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/17/877717/2/check/DellEMC_PowerFlex/8b0483e/DellEMC_PowerFlex/1756 : FAILURE in 1h 45m 36s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"14e2e8179c8196d8506800d43b2daa03e13bf01b","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2023-03-17 17:29:59.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/17/877717/2/check/DellEMC_SC_FC/8bdac3d/DellEMC_SC_FC/1222 : SUCCESS in 1h 04m 04s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/17/877717/2/check/DellEMC_SC_iSCSI/fa46e2f/DellEMC_SC_iSCSI/1202 : SUCCESS in 1h 55m 02s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"20a9f6dd4b9257b7d4ba09923ee52f512f92bf6b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-17 18:42:23.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/17/877717/2/check/iscsi-driver-alletra-6k/6e68019/6e68019.html : FAILURE in 2h 11m 44s","accounts_in_message":[],"_revision_number":2},{"id":"41f4667fe213572bfd6dce1321c5a779d99774a1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-17 18:46:01.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-17-877717-2/ns5_nfs : SUCCESS in 3h 28m 37s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-17-877717-2/ns5_iscsi : SUCCESS in 3h 27m 8s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"d0f3fad53b29c4cacf0b93afeeb3414d0ab6e951","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-03-17 19:34:50.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0bcc8aba9213846924de109d8a4d054ef9f2ca2c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-17 19:41:56.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/17/877717/2/check/DellEMC_Unity_FC/98060cf/DellEMC_Unity_FC/2522 : SUCCESS in 1h 29m 59s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/17/877717/2/check/DellEMC_Unity_iSCSI/830554f/DellEMC_Unity_iSCSI/2370 : SUCCESS in 2h 04m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"2974fce03a1888a424ec02493043c95c033b7bb1","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-03-17 20:05:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2d62debc8f6d38d335ff554f939506533f946d42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 20:06:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0c0d28f7d7b348f3c62f5c6aad3ddd615992b6cf","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-03-17 21:01:41.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/877717-2/HPMSA-iSCSI/ : SUCCESS in 1:39:08\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/877717-2/HPMSA-FC/ : SUCCESS in 1:19:15","accounts_in_message":[],"_revision_number":2},{"id":"d444d64176a4425953d0690e9e4ba286c0d68f03","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-03-17 22:23:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/877717-2/PVME-iSCSI/ : SUCCESS in 1:50:02\n- PVME-FC https://os-logs.seagate.com/dell-ci/877717-2/PVME-FC/ : SUCCESS in 1:33:53","accounts_in_message":[],"_revision_number":2},{"id":"b4a9f4cfb37b163c8754d9d3de6a92533c629653","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-03-17 22:24:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/877717-2/Seagate-iSCSI/ : SUCCESS in 1:47:22\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/877717-2/Seagate-FC/ : SUCCESS in 1:34:02","accounts_in_message":[],"_revision_number":2},{"id":"8310fc28c55dcd18503f20e80d224c93d776544c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 23:31:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a16c6a2cf5e249188eb6b866634f1a83\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/560300fe0f164d1bb0ef287870b3de93 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4ccae29c3c440cc9ea29d80e4d3ee6f : SUCCESS in 12m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ca20b7c93ef4d6ab7cf41b0b0533c8a : SUCCESS in 12m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/073a1b504352487a81fbfab64be822c0 : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":2},{"id":"e0da70f7482279568e48b0b4a23c32ae2e57c30b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 23:31:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1bb8150aa227561f71ef89250886ad3e6f6fd96e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-17 23:32:00.000000000","message":"Patch Set 2:\n\nBuild Started http://openstack-ci.spb.yadro.com:8080/job/openstack-gerrit-trigger-fc/145/ (1/2)","accounts_in_message":[],"_revision_number":2},{"id":"703d3694e0f25037a17438c7bf620ed5628830a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 23:33:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fe061e9812b48a8ba3f495f4214ce93\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f393244a70814906ad61e51d28e00afa : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2},{"id":"1e673cb7e182b59f384e9be7a03648a89d593d85","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-17 23:38:51.000000000","message":"Patch Set 2:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/1207 \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2},{"id":"18603d74a56bc4ac8f3d2ba1e5c634f5a28e500a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-19 23:40:16.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/17/877717/2/check/iscsi-driver-nimble/5b04a86/5b04a86.html : FAILURE in 1h 02m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d7cf4a89abf68404bb32f0798fe590e846a62164","revisions":{"bb3de1db4b1672d03977db667eca448014d5c859":{"kind":"REWORK","_number":1,"created":"2023-03-16 20:40:18.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/17/877717/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/877717/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/877717/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/877717/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/877717/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/877717/1"}}},"commit":{"parents":[{"commit":"42d5d1d6487c7b838e7c01bd96757d6c59116fc6","subject":"Merge \"PowerMax Driver - Fix for force flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/42d5d1d6487c7b838e7c01bd96757d6c59116fc6"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2023-03-16 20:37:09.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2023-03-16 20:39:55.000000000","tz":-300},"subject":"Make PowerMax tests stable","message":"Make PowerMax tests stable\n\nThe problem manifests itself as an unstable result of\ntest_check_force. It fails at Zuul once in a while. But this\nseems impossible: there is no randomness or timing component.\n\nBut apparently someone forgot a deepcopy().\n\nStill, we also make test_check_force not to depend on the\nglobal specs, seems like an obvious oversight.\n\nChange-Id: If1ef9bceae96ef044209c758eee2fd22b44efa49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bb3de1db4b1672d03977db667eca448014d5c859"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/bb3de1db4b1672d03977db667eca448014d5c859"}]},"branch":"refs/heads/master"},"d7cf4a89abf68404bb32f0798fe590e846a62164":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-03-17 15:16:52.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/17/877717/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/17/877717/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/877717/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/877717/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/17/877717/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/17/877717/2"}}},"commit":{"parents":[{"commit":"42d5d1d6487c7b838e7c01bd96757d6c59116fc6","subject":"Merge \"PowerMax Driver - Fix for force flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/42d5d1d6487c7b838e7c01bd96757d6c59116fc6"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2023-03-16 20:37:09.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2023-03-17 15:16:52.000000000","tz":0},"subject":"Make PowerMax tests stable","message":"Make PowerMax tests stable\n\nThe problem manifests itself as an unstable result of\ntest_check_force. It fails at Zuul once in a while. But this\nseems impossible: there is no randomness or timing component\nin this test. The randomness occurs because of cross-test\ninterference.\n\nThe root cause is that apparently someone forgot a deepcopy().\n\nWe also make test_check_force not to depend on the global specs,\nseems like an obvious oversight. It is not the fix, only tidies\nthe code.\n\nChange-Id: If1ef9bceae96ef044209c758eee2fd22b44efa49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7cf4a89abf68404bb32f0798fe590e846a62164"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7cf4a89abf68404bb32f0798fe590e846a62164"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
