)]}'
{"id":"openstack%2Fcinder~877799","triplet_id":"openstack%2Fcinder~stable%2F2023.1~I54b81a568a01af44e3f74bcac55e823cdae9bfbf","project":"openstack/cinder","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2023-03-18 07:27:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I54b81a568a01af44e3f74bcac55e823cdae9bfbf","subject":"Restore into sparse volumes","status":"MERGED","created":"2023-03-17 14:24:46.000000000","updated":"2023-04-22 15:27:51.000000000","submitted":"2023-03-18 07:27:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877799","meta_rev_id":"6173df6ccb52505123c532cc1b316fd7ad18ad8e","_number":877799,"virtual_id_number":877799,"owner":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-18 07:27:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2023-03-18 03:59:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2023-03-17 15:09:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-17 14:29:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2023-03-18 03:59:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-17 14:24:46.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2023-03-17 14:29:23.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-03-17 14:36:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-17 14:37:49.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-17 15:09:03.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2023-03-17 17:12:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-17 17:30:14.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-03-17 17:47:15.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-03-17 18:05:49.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-03-18 03:59:17.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"e7cb003e7318b6161033d832c2a9b966f0494f71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2023-03-17 14:24:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"279d4524a69818af1634c1a46442762790fec7ee","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-03-17 14:29:23.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"635bf008a3eed52b4d5588e56f460ad946130cde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 14:36:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ce9fe9bb08d4065bd668ec283477c4a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3269a516f4564b48b7853e4224b0132c : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4e062ba2ed7e4577afdeebfda70719a9 : SUCCESS in 9m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7760222fcbf36cb8bc5db929f53cfecccc769dd7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-17 14:37:49.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder iscsi-driver-primera-cinder : RETRY_LIMIT in 0s","accounts_in_message":[],"_revision_number":1},{"id":"90500a266f653120f80b9db727bef046be9b0a7d","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-03-17 15:09:03.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"89eb824102c9238d7a8e441449e4396333899c65","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-17 15:16:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/877799 : FAILURE in 25m 08s","accounts_in_message":[],"_revision_number":1},{"id":"f18f8c61fd8487df8a86aeba0cc463ab8c3c45de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 17:12:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35b291a346584ba980c8504cbcdc63d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83de0fdd113a4227909274beac914d61 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/deaf378e7842405f8efa37c7424479af : SUCCESS in 10m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c78efa2e8254897b2afbf06e4f8641c : SUCCESS in 10m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3851726a90904ecc9d9e96463a61794d : SUCCESS in 12m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/604426cdd85d4521a0c2fa5dc05e3144 : SUCCESS in 1h 36m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00989fd813334cd99e6b61ede906b785 : SUCCESS in 58m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcd6ff6f44864444a560d6a11fa99f53 : SUCCESS in 8m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/cf09fcc08b7d4228b21beb280557106b : SUCCESS in 15m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/78afdbd8bad64c2f9876847c006efca8 : SUCCESS in 5m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1125dfdeb550419fb769cccf77d09e79 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/05126a1109af446fa5aa38d417901296 : SUCCESS in 5m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b656866d7c5e481fb2fea6e056240fe7 : SUCCESS in 6m 02s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4ff16497278349a5b1d03bef397cf038 : SUCCESS in 58m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b5970ae74cbb4145a85fe721f2f21808 : FAILURE in 5m 39s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7ffaffb307644cd2b77d18a4f71722a8 : SUCCESS in 1h 42m 06s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6eb0bb7874d34cdea531451d4c8e5456 : FAILURE in 1h 27m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/791b36e4ad034f549a162bcc830efb87 : SUCCESS in 1h 38m 59s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e03fa279c6be4579abe142106677398b : SUCCESS in 1h 18m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/bc8cc6dc7b99492a86979d526938af7b : SUCCESS in 47m 29s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/b6981b4135d14c2da09c15b942615876 : SUCCESS in 1h 11m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9ff0fb738c8244c2b000d1d77815d89b : SUCCESS in 2h 30m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da5761d95cc0490c8bd670d3ebcb916f : SUCCESS in 59m 06s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba843589de964b54a5e1b78cd5753142 : SUCCESS in 1h 01m 15s","accounts_in_message":[],"_revision_number":1},{"id":"2e291908bf755c317ff7345a38ab1881a9466265","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-03-17 17:30:14.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_03_2023_17_22_10\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"b916b9ac819f2d17031b50bf411b5c4082542f42","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-17 17:47:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/99/877799/1/check/iscsi-driver-alletra-6k/bbe3c9b/bbe3c9b.html : FAILURE in 1h 32m 52s","accounts_in_message":[],"_revision_number":1},{"id":"def21ebc93db81cb1d3a8de5886ef793e856736f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-17 18:05:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/99/877799/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/41d56e6/ : SUCCESS in 1h 52m 42s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/99/877799/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/396dcfd/ : FAILURE in 1h 53m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a247f20808758444831109f637194f72324b8fd0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-03-18 03:59:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"11d1f5ab7ff7a20e58d5d2926b1161fa22bf1f11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-18 04:00:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a156e57c23cfbc097c47ee7b4a0291d28bf5d8be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-18 07:27:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f95bd499c6464df48da5e4cc19b6433f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d55249fa5480460fbc91724d5a14c4f9 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c60ddf4c5fc74d16b8251b0c362522f0 : SUCCESS in 11m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f2f7d1ca95142489c225a6c728063c1 : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53b22491f3fb4deba429d1ec3e6cd32a : SUCCESS in 12m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/51c3ce2fc9e54e9b923514fc5a427d09 : SUCCESS in 1h 04m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/13268196803744fa98d3f55d0a724425 : SUCCESS in 1h 25m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ca3f3edda3645e6b0e50450f71f0fb5 : SUCCESS in 1h 03m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d32f70f5ada643bb8450ffa463064a82 : SUCCESS in 8m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/df70d807d655477c814d954260e3644d : SUCCESS in 1h 24m 10s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1f326a817f194a62b2b2d952f47fc4be : SUCCESS in 1h 23m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72c7ddd7f4694757896f013da35b7338 : SUCCESS in 53m 41s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c1bea707f694d868f2890831869788b : SUCCESS in 1h 26m 43s","accounts_in_message":[],"_revision_number":1},{"id":"28e3f60978c6fa52ac0b320a014b1b44ab232bc3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-18 07:27:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0b728d3eb2159ef0e0372f3b0e3b5a202a1bbedf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-18 07:28:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae5ab9848de1427b903799b266628c0a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/180fda35ccd349b9b6b904d1abc6a9d1 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/273ef442e58b4cf3bc38163c8711819d : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1},{"id":"6173df6ccb52505123c532cc1b316fd7ad18ad8e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-22 15:27:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/99/877799/1/check/iscsi-driver-nimble/f30f9df/f30f9df.html : FAILURE in 31m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c56b79caa3958dec47f6aa1815abc7638a42952a","revisions":{"c56b79caa3958dec47f6aa1815abc7638a42952a":{"kind":"REWORK","_number":1,"created":"2023-03-17 14:24:46.000000000","uploader":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"ref":"refs/changes/99/877799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/99/877799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/99/877799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/99/877799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/99/877799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/99/877799/1"}}},"commit":{"parents":[{"commit":"7ee5824c64b1c3c85d1ce1636bdecd86acb64970","subject":"Add 2023.1 (Antelope) prelude","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/7ee5824c64b1c3c85d1ce1636bdecd86acb64970"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2022-07-20 17:15:21.000000000","tz":-300},"committer":{"name":"Jon Bernard","email":"jobernar@redhat.com","date":"2023-03-17 14:24:44.000000000","tz":-240},"subject":"Restore into sparse volumes","message":"Restore into sparse volumes\n\nWe take the approach to detect zero data when restoring and not\nwriting it out to the volume. This works exactly as expected\nwhen restoring into Linux LVM volumes and into Ceph RBD volumes,\nprovided they are freshly created.\n\nCloses-bug: #2007615\nChange-Id: I54b81a568a01af44e3f74bcac55e823cdae9bfbf\n(cherry picked from commit 477b1342747ddeb3ab9a586b33c06f1e51f37df8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c56b79caa3958dec47f6aa1815abc7638a42952a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/c56b79caa3958dec47f6aa1815abc7638a42952a"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
