)]}'
{"id":"openstack%2Fcinder~877826","triplet_id":"openstack%2Fcinder~master~I6dacbb2aba37dac27ad741aee01cd4f1355198cb","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2023-03-17 23:31:57.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2023-03-17 23:31:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6dacbb2aba37dac27ad741aee01cd4f1355198cb","subject":"Fix: PowerMax test with flipping force flag","status":"MERGED","created":"2023-03-17 16:46:38.000000000","updated":"2023-04-23 00:54:34.000000000","submitted":"2023-03-17 23:31:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877826","meta_rev_id":"45b961009a490dad4522982b2ef8efd54937569c","_number":877826,"virtual_id_number":877826,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-17 23:31:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2023-03-17 20:23:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2023-03-17 19:47:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2023-03-17 20:23:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-17 16:46:45.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-03-17 17:02:02.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-17 17:07:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-17 17:22:08.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-03-17 18:11:30.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-17 18:21:10.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-03-17 19:18:46.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-03-17 19:47:16.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2023-03-17 20:15:09.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-17 21:04:55.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-03-17 22:36:41.000000000","updated_by":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"reviewer":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"state":"CC"},{"updated":"2023-03-17 22:38:38.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-17 23:40:02.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-03-18 01:42:48.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-03-18 04:39:30.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-03-18 06:57:27.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2023-03-18 06:57:27.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"}],"messages":[{"id":"893f3faef8253bdb011c5c2a51d395b577232a0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-03-17 16:46:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"255af6b12ef14d813cf399bd1293c1906f47b3f0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-17 16:46:45.000000000","message":"Patch Set 1:\n\nBuild Started http://openstack-ci.spb.yadro.com:8080/job/openstack-gerrit-trigger-fc/139/ (1/2)","accounts_in_message":[],"_revision_number":1},{"id":"fcffd66526d515318d973e42463db683bc791776","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-17 16:53:28.000000000","message":"Patch Set 1:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/1201","accounts_in_message":[],"_revision_number":1},{"id":"5b5db28b64eca6cd9884ab769c6ff2e565d2b522","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-17 17:02:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder finger://openstack-ci-fc/eefbae3d3b324926aab20d1f16dd1039 : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":1},{"id":"36bff11fecdde93d0c7b68f655a8fa6195298933","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 17:07:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1dd5f639ec345738ee60353690f85fc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d9e5ec29ccd41458b8a0183503c0018 : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e436cb286ca24c24a2f06a55af5071e5 : SUCCESS in 11m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41a7728bf76e4474b3ab35fc38b10635 : SUCCESS in 10m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/188367b7bbb54b42af723eafb67ca646 : SUCCESS in 13m 33s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9179f0e8158a4626b846c6e4ac5108b6 : FAILURE in 13m 07s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/bf90ab854f35409eac82b14652219770 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"05c0ff7cd99626610d1e678d709d5ce5148782ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 17:09:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89cdeadbfe4e41adb7dc3e5e250e67ec\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/55aad9b899ba4c04a57a6607a57649de : SUCCESS in 22m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4e0cb3857cd843f8a641f575dda45d15 : FAILURE in 11m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"801fd14ceff544f458805e6adea839e8d91ebc80","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-17 17:16:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/877826 : FAILURE in 27m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d401f22c5f496054a761148e73990b2e8eab22b0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-03-17 17:22:08.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_03_2023_17_14_07\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"f2327c5e35b2a400022d5e4ba38f19e4b850657e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-17 18:11:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/34d80162534d4e69a83904f1f114eb58 : SUCCESS in 1h 22m 51s","accounts_in_message":[],"_revision_number":1},{"id":"830425fd93329e5662ecfdd5091549e747bba59a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-03-17 18:21:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c69aa7d373a23049424e9762c8e606d8af175fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 18:21:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"49904c376eae4b004c5c7bc302df2f7369d53764","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 18:38:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/08917471670743fb9f0c9223bb3fdc83\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/316f952dcbc743ad8c819947593c3c31 : SUCCESS in 8m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7c86906c7d94211b5379de01ef072aa : FAILURE in 12m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ea4bda6b1824c74ad6bcb5ec8c44de0 : SUCCESS in 13m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d70a9cbc09b48b09e89d810050b5c55 : SUCCESS in 16m 03s","accounts_in_message":[],"_revision_number":1},{"id":"950aaeec248c5740c7e1ee4309842645fa058029","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-17 19:18:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/26/877826/1/check/iscsi-driver-alletra-6k/5848e09/5848e09.html : FAILURE in 1h 27m 17s","accounts_in_message":[],"_revision_number":1},{"id":"29d7e4fe80cacf9f1570b15221f680d198c823c2","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-03-17 19:30:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9bf4cadfe5acb4eda82103dac1d025e6973f74f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-03-17 19:34:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8941bd3e507242ae15a5a9c23e93da3b5bcf4500","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-17 19:34:31.000000000","message":"Patch Set 2:\n\nBuild Started http://openstack-ci.spb.yadro.com:8080/job/openstack-gerrit-trigger-fc/142/ (2/2)","accounts_in_message":[],"_revision_number":2},{"id":"deeb4e8ad9c9153030f38af6f913865a9df91f30","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-17 19:41:03.000000000","message":"Patch Set 2:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/1204","accounts_in_message":[],"_revision_number":2},{"id":"f1d0deda74fca3e7fdbbf8d4f6b232fc51971d8c","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-03-17 19:47:16.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"569cfa23ae44b71a1275d60919eab9b7f2ccca6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 19:48:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0034553bfcd0469db19ffd298195b2b8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4bfc7eaefaff44c5addd15edc5d8c58d : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f8586fe0b16e4ceca4262c5848985faf : SUCCESS in 13m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1e24bf9598d61f09fe5e62d03a0ba6cd4e1e2fad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 19:59:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ae20088695b4c4cb6e5e0993bc8f74c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34840c0dccc94cf481bf7cca31aebd10 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01f483bb32d14f5bb307db86c499aa01 : SUCCESS in 11m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/567a5662508e426dacc56f7a3df945ba : SUCCESS in 10m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44c33fb5f7e94b719d0680908b815d71 : SUCCESS in 12m 22s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/583cf9fc219b424892f4bb5592187169 : SUCCESS in 17m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/934313f598d243c98ef87342fe4dbca5 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":2},{"id":"2c05fd476eb8163c3cf461209ee60a2ff7388335","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-17 20:03:37.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/877826 : FAILURE in 26m 46s","accounts_in_message":[],"_revision_number":2},{"id":"e2cf661fe1cd5b1997eac764f30ec086625b62cd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-03-17 20:07:50.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/17_03_2023_19_59_47\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"9806fd34b4d0a869f1174e634dca15dffee1a652","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-17 20:15:09.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-26-877826-1/ns5_nfs : SUCCESS in 3h 24m 32s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-26-877826-1/ns5_iscsi : FAILURE in 3h 28m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"f82e5f148007fcd58baa33feb457f6ce834ebb18","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-03-17 20:19:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dcba02ec56503e790ed68b8b22e87f9639d8280c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-03-17 20:23:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5bfb479c417f5ff4c17265d46c7173f2e29fb50c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 20:24:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"48ba26b1bdc72400c481f2086c0ff1c7d0313283","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-17 20:26:32.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder finger://openstack-ci-fc/5ed31e9aeed849d3b86b1b18087ca59e : RETRY_LIMIT in 37s","accounts_in_message":[],"_revision_number":2},{"id":"b8e95c86a3ff3cd07b0e6cbbee88ac7e655c4ab2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-17 20:53:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/c633fc6d2e52481abb648279022c07fb : SUCCESS in 1h 16m 52s","accounts_in_message":[],"_revision_number":2},{"id":"b0f63acb759b0adb2ae8ec0d305ce79b7608202b","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-17 21:04:55.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/3f1bd09dd9224a2695a9bb4cbb070484 : SUCCESS in 1h 29m 47s","accounts_in_message":[],"_revision_number":2},{"id":"0ac687147bfac8b10ba366a508464bdcce61ae85","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-17 21:25:11.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/26/877826/2/check/iscsi-driver-alletra-6k/dd47470/dd47470.html : FAILURE in 1h 49m 49s","accounts_in_message":[],"_revision_number":2},{"id":"6821d732774989f8dbb19a4f76726ab4eb14aad2","author":{"_account_id":25678,"name":"DellEMC SC CI","email":"dell.emc.sc.series.ci@emc.com","username":"dell.emc.sc.series.ci"},"date":"2023-03-17 22:36:41.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_SC_FC https://elab-os-logsrv.delllabs.net/26/877826/2/check/DellEMC_SC_FC/849f5fd/DellEMC_SC_FC/1226 : SUCCESS in 1h 04m 19s\n- DellEMC_SC_iSCSI https://elab-os-logsrv.delllabs.net/26/877826/2/check/DellEMC_SC_iSCSI/f52bc3c/DellEMC_SC_iSCSI/1206 : SUCCESS in 1h 35m 17s\n\nLeave a comment with \u0027run-DellEMC SC CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_SC_CI","accounts_in_message":[],"_revision_number":2},{"id":"396d797fcf0e287b9732194d7e511f8b19480850","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-03-17 22:38:38.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/26/877826/2/check/DellEMC_PowerFlex/a752e0b/DellEMC_PowerFlex/1760 : SUCCESS in 1h 34m 50s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"7b68c28f61baffa757fc3d19be48f09b07b14e9d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-17 23:02:49.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-26-877826-2/ns5_nfs : SUCCESS in 3h 25m 57s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-26-877826-2/ns5_iscsi : SUCCESS in 3h 27m 56s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"8a95197ff11c9b911846420a2e87f1a2f4bf510c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 23:31:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aab5b8f6ea6e42c296ce56b1c7f43ab4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89d83d27ee8a4fc68687e053bd36f96c : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2a3f41166e347a0b8baf7472bdf5ea4 : SUCCESS in 10m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58ab54df44cc488dba7d0e1f8790e032 : SUCCESS in 13m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad3a9ac380e94bc584911f8a2eeb5019 : SUCCESS in 14m 03s","accounts_in_message":[],"_revision_number":2},{"id":"f4a79517bcd1aacb07b731a7a2cfca0d254f5be0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 23:31:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d3408f27fe7760f41a6295ce3590d7b1a3ef7a44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 23:34:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28d49b3ec83a42f8854915ad023ef99b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/606af04b58314e85b38b2326a7eab42e : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2},{"id":"7fe2fee443e0e703746631e3fc68418fe5c0c301","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-17 23:38:52.000000000","message":"Patch Set 2:\n\nBuild Started http://openstack-ci.spb.yadro.com:8080/job/openstack-gerrit-trigger-fc/146/ (2/2)","accounts_in_message":[],"_revision_number":2},{"id":"b0c8d8e68885091d469319fe5aaaca6dbc410f5b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-17 23:40:02.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/26/877826/2/check/DellEMC_Unity_FC/8d16348/DellEMC_Unity_FC/2525 : SUCCESS in 1h 30m 34s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/26/877826/2/check/DellEMC_Unity_iSCSI/4ccef2f/DellEMC_Unity_iSCSI/2372 : SUCCESS in 2h 02m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"ad1dad352de5e2e43ad75b431b91db9075b3b218","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-17 23:45:11.000000000","message":"Patch Set 2:\n\nFAILED build log link http://tempest-openstack.spb.yadro.com/gerrit-trigger/1208 \u0027   --tag autogenerated:jenkins-gerrit-trigger","accounts_in_message":[],"_revision_number":2},{"id":"8ec33add6dde78e48bff96c145f08a683fd8134c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-03-18 01:42:48.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/26/877826/2/check/DellEMC_PowerStore_iSCSI/491328b/DellEMC_PowerStore_iSCSI/1153 : FAILURE in 1h 14m 52s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/26/877826/2/check/DellEMC_PowerStore_FC/da918fa/DellEMC_PowerStore_FC/1275 : FAILURE in 1h 29m 09s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/26/877826/2/check/DellEMC_PowerStore_NFS/c772a8f/DellEMC_PowerStore_NFS/792 : FAILURE in 1h 21m 31s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/26/877826/2/check/DellEMC_PowerStore_NVMeTCP/be4d677/DellEMC_PowerStore_NVMeTCP/655 : FAILURE in 2h 51m 31s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"977de6191e0db254c803cc26bf3fe3b987a14ce8","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-03-18 04:39:30.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/877826-1/HPMSA-iSCSI/ : SUCCESS in 1:37:12\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/877826-1/HPMSA-FC/ : SUCCESS in 1:19:02","accounts_in_message":[],"_revision_number":1},{"id":"7fa4e5dfe7eed5f82270aa49408fb9a682bf0cfe","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-03-18 06:57:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/877826-1/Seagate-iSCSI/ : SUCCESS in 1:52:04\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/877826-1/Seagate-FC/ : SUCCESS in 1:36:02","accounts_in_message":[],"_revision_number":1},{"id":"fd57c7eac1f05311674f902d283d0cbafb22c503","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-03-18 06:57:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/877826-1/PVME-iSCSI/ : SUCCESS in 1:52:25\n- PVME-FC https://os-logs.seagate.com/dell-ci/877826-1/PVME-FC/ : SUCCESS in 1:35:37","accounts_in_message":[],"_revision_number":1},{"id":"4570b6d3ea2002e4528beef4ca6c98bc2e57ae75","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-03-18 12:09:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/877826-2/HPMSA-iSCSI/ : SUCCESS in 1:36:50\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/877826-2/HPMSA-FC/ : SUCCESS in 1:17:30","accounts_in_message":[],"_revision_number":2},{"id":"9c1b4c074889d25fbc373893fd8f034cd68460b5","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-03-18 15:29:09.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/877826-2/PVME-iSCSI/ : SUCCESS in 1:49:44\n- PVME-FC https://os-logs.seagate.com/dell-ci/877826-2/PVME-FC/ : SUCCESS in 1:34:52","accounts_in_message":[],"_revision_number":2},{"id":"28003d207fb6c2dc71a23c7fcc27c27c46b5b82a","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-03-18 15:31:55.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/877826-2/Seagate-iSCSI/ : SUCCESS in 1:51:29\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/877826-2/Seagate-FC/ : SUCCESS in 1:34:42","accounts_in_message":[],"_revision_number":2},{"id":"45b961009a490dad4522982b2ef8efd54937569c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-23 00:54:34.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/26/877826/2/check/iscsi-driver-nimble/d752ebc/d752ebc.html : FAILURE in 31m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b466bb3c1ba1eb0c7431fff3fa8239aec0ce6ebc","revisions":{"fccd6b347e35c8a08163cf7450d5cd9ec3fad014":{"kind":"REWORK","_number":1,"created":"2023-03-17 16:46:38.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/26/877826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/26/877826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/877826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/877826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/877826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/26/877826/1"}}},"commit":{"parents":[{"commit":"42d5d1d6487c7b838e7c01bd96757d6c59116fc6","subject":"Merge \"PowerMax Driver - Fix for force flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/42d5d1d6487c7b838e7c01bd96757d6c59116fc6"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2023-03-17 16:43:09.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2023-03-17 16:43:09.000000000","tz":0},"subject":"Fix: PowerMax test with flipping force flag","message":"Fix: PowerMax test with flipping force flag\n\nThis patch fixes the error faced in a recent gate run[1].\nThis is a random behavior but there are other places where\nsimilar change is required.\nSince the whole powermax unit tests need to be refactored to do\nthis correctly, this patch only addresses the failure faced in\nthe gate run.\n\n[1] https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_bd9/852654/16/gate/openstack-tox-py310/bd9f0d9/testr_results.html\n\nChange-Id: I6dacbb2aba37dac27ad741aee01cd4f1355198cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fccd6b347e35c8a08163cf7450d5cd9ec3fad014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fccd6b347e35c8a08163cf7450d5cd9ec3fad014"}]},"branch":"refs/heads/master"},"b466bb3c1ba1eb0c7431fff3fa8239aec0ce6ebc":{"kind":"REWORK","_number":2,"created":"2023-03-17 19:34:21.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/26/877826/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/26/877826/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/877826/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/877826/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/26/877826/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/26/877826/2"}}},"commit":{"parents":[{"commit":"42d5d1d6487c7b838e7c01bd96757d6c59116fc6","subject":"Merge \"PowerMax Driver - Fix for force flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/42d5d1d6487c7b838e7c01bd96757d6c59116fc6"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2023-03-17 16:43:09.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2023-03-17 19:33:30.000000000","tz":0},"subject":"Fix: PowerMax test with flipping force flag","message":"Fix: PowerMax test with flipping force flag\n\nThis patch fixes the error faced in a recent gate run[1].\nThis is a random behavior but there are other places where\nsimilar change is required.\nSince the whole powermax unit tests need to be refactored to do\nthis correctly, this patch only addresses the failure faced in\nthe gate run.\n\n[1] https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_bd9/852654/16/gate/openstack-tox-py310/bd9f0d9/testr_results.html\n\nChange-Id: I6dacbb2aba37dac27ad741aee01cd4f1355198cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b466bb3c1ba1eb0c7431fff3fa8239aec0ce6ebc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b466bb3c1ba1eb0c7431fff3fa8239aec0ce6ebc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
