)]}'
{"id":"openstack%2Fcinder~878168","triplet_id":"openstack%2Fcinder~stable%2Fzed~I739ae6ae5b4d9b074d610f6a70371b294a3e70f8","project":"openstack/cinder","branch":"stable/zed","topic":"rbd_secret_uuid-stable/zed","attention_set":{},"removed_from_attention_set":{"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2023-04-12 19:47:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I739ae6ae5b4d9b074d610f6a70371b294a3e70f8","subject":"RBD: Default rbd_secret_uuid to the cluster FSID","status":"MERGED","created":"2023-03-22 13:28:18.000000000","updated":"2023-04-12 19:49:01.000000000","submitted":"2023-04-12 19:47:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878168-rbd_secret_uuid-stable/zed","cherry_pick_of_change":860090,"cherry_pick_of_patch_set":1,"meta_rev_id":"5d8f078c3e2900b43afbb20f1a53111146857610","_number":878168,"virtual_id_number":878168,"owner":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-12 19:47:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"recommended":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"all":[{"value":1,"date":"2023-03-28 09:52:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},{"value":1,"date":"2023-03-22 21:49:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-12 17:44:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":2,"date":"2023-04-12 18:05:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":1,"date":"2023-04-12 18:05:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-22 13:28:18.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"CC"},{"updated":"2023-03-22 13:30:17.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-22 13:48:00.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"CC"},{"updated":"2023-03-22 13:56:27.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-22 13:58:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-22 15:18:01.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-03-22 15:46:02.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-03-22 16:00:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-22 16:08:17.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-03-22 21:49:07.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2023-03-24 21:49:06.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-03-25 20:07:07.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-28 09:52:14.000000000","updated_by":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"reviewer":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"state":"REVIEWER"},{"updated":"2023-04-12 17:44:36.000000000","updated_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2023-04-12 18:05:53.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"19ed60277a7cca19f43ae6cd4fdc1f3f43c5ed1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2023-03-22 13:28:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"54f423ae0a1eaa755a361e2e8c2eda6b5143a66d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-22 13:30:17.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-68-878168-1/ns5_nfs : FAILURE in 1m 25s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-68-878168-1/ns5_iscsi : FAILURE in 1m 22s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"c730daac20f7c461f19b70ea1370969b45ae63ac","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-22 13:56:27.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/878168 : FAILURE in 26m 36s","accounts_in_message":[],"_revision_number":1},{"id":"087772e67191762d1008826620a60d5022b78b12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 13:58:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb6ed28e4cb4499499f606f378e77afc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/41aeb1f731264efdb7c67461d2d7bf98 : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/89585bba13504d7c9b8aa610f7e15560 : SUCCESS in 27m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"54d4e44b0bc40395178c360cba9f2319d714c2ee","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-22 15:18:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/68/878168/1/check/iscsi-driver-alletra-6k/ad43e40/ad43e40.html : FAILURE in 1h 45m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9de9e71ee2e54a07663002d5770662cfbc7ee4f0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-22 15:46:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/68/878168/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/0d5b920/ : SUCCESS in 1h 40m 49s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/68/878168/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/91a7065/ : FAILURE in 2h 01m 47s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/68/878168/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/6c91ef3/ : SUCCESS in 1h 28m 07s\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/68/878168/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/fafd4ba/ : FAILURE in 2h 16m 26s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/68/878168/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/38faf08/ : SUCCESS in 1h 56m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"596b9f4aa461d39fcd5c71d72e52b644094643b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 16:00:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae98f98cb01d4f94a3c24eedd3185326\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad8e1fe65d3d4f3bb3857b4f010e7321 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb267b604f244f12a5b769f8f72bb93d : SUCCESS in 11m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8dc6e697b96d4779b26015391185ca5c : SUCCESS in 12m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30a79bb4e335441694e96d8c182f5937 : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9e97695b7524339ad430de04f50ec50 : SUCCESS in 11m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/04f1fa9cff014e01ae614fe7bceb6681 : SUCCESS in 1h 04m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5be72b23c47847de9750779730365dca : SUCCESS in 1h 15m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7cc4e9b28188454eb7bbcb3fbbf6ae0d : SUCCESS in 53m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d61a534d7b64e488343a61a8d8c44ce : SUCCESS in 8m 27s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/b3b4062456ef41548605191366c7b6a2 : SUCCESS in 14m 41s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/32f80390e4fa44fea849c2168a3a330d : SUCCESS in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e51b6db3575d4b2cb6307c51b1a95af5 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/67cc7b466d0d46b8974c1408a41aa22d : SUCCESS in 5m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/032fb77d6ab84ce6a06dd43ba4ec6758 : SUCCESS in 6m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f68daa4691754051b774828fdb371021 : SUCCESS in 46m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/69a310b42ff24ff0be3ae916cff371bb : FAILURE in 5m 15s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/12730595bd3e4880b3d421588c8d0488 : SUCCESS in 1h 34m 59s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f3cdce293ae34e4abdb9efb527d89044 : FAILURE in 1h 35m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8d219c6ccdf440acad9e88c02425f575 : SUCCESS in 1h 20m 27s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1edcd95e9c044744ad5a2b7fe9ed4a4b : SUCCESS in 1h 12m 03s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/bb4f0e4eda9a4bd7873fa9a1dec66637 : SUCCESS in 1h 42m 03s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7670b75a2240435698bd607f27542360 : SUCCESS in 36m 17s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5b0eebda86eb4cf9a7f8ef9b4fac38b1 : SUCCESS in 1h 19m 37s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/39fc53da22ab47248609024c8bcbbcad : SUCCESS in 2h 25m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/acbf51d247cb45f3a68cb8e41c6f340a : SUCCESS in 45m 11s","accounts_in_message":[],"_revision_number":1},{"id":"4cd72b198898ad24788ed8a9e999fe5e1f1d8e53","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-03-22 16:08:17.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/22_03_2023_16_00_28\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"bb3335777df5f8b7d923ea45f9da2ca4aa31eab9","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2023-03-22 21:49:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a60ae9e0b741959c634ce3def4a5a40a20c95fa9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-24 03:50:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder finger://openstack-ci-fc/f55eb908636f4d3486947239bb8672a6 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":1},{"id":"e8e1e375c684bc87363bb493622e42e321cb726f","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-03-24 21:49:06.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/68/878168/1/20230325010125/NEC-V-ISCSI : SUCCESS in 2h 44m 06s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/68/878168/1/20230325010125/NEC-V-FC : FAILURE in 5h 47m 38s","accounts_in_message":[],"_revision_number":1},{"id":"094d77daa81f228aebfc330e3dcd5e51ea2fa580","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-03-25 20:07:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/68/878168/1/ : SUCCESS in 1h 12m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ac904b21c458d481ff480b3605df9bc71834c9aa","author":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"date":"2023-03-28 09:52:14.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b519542702ca47ca4dc4dde34a06a7cc86c64aaf","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2023-04-12 17:44:36.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a67e140846a034646b749af578bebef9948276fa","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-04-12 18:05:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"86e4d07f2d6bee5eb088964c71e31e52777f3f23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 18:06:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d718e83e4dabe140808537245a3b5e4a51ffb9ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 19:47:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f90db78d51e84f309032bad9df86b3ee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84b5b2f09a8041abbf47ceced55d4a84 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08cf590d8b75457f95c2a462b303b1bd : SUCCESS in 10m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf0536a23f4947afbcd789705db25502 : SUCCESS in 10m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/230337eb452c4cf69a69517359bf9018 : SUCCESS in 13m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/adb3f4675178431dba113009396e6c50 : SUCCESS in 1h 05m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9b67acc966c94fc6b4b50c3a5fe6f817 : SUCCESS in 1h 32m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7cc0b05f84b740b7a2614c2fbc32020c : SUCCESS in 44m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/422a1585e8d64feea6e4a2832d790302 : SUCCESS in 8m 47s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2d069721092741c3a0cad6d897d38f3c : SUCCESS in 1h 08m 10s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/240d58859216455ab3b670ff6315b104 : SUCCESS in 1h 11m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6931dfe2b99f46009f20d240c3ea4b64 : SUCCESS in 57m 57s","accounts_in_message":[],"_revision_number":1},{"id":"56e90fb3d649ee2cf84997f8b5aaf14b92be25c6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 19:47:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5d8f078c3e2900b43afbb20f1a53111146857610","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 19:49:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/387ab91600e24bbcb42fed63f0a62736\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1ef7138439b4b879be11444c58fea09 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4aaf059fd83c4a45b29f990ef1e79bc6 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30b70ac5a07daf8feb82c8bc51a579b0fd0b591f","revisions":{"30b70ac5a07daf8feb82c8bc51a579b0fd0b591f":{"kind":"REWORK","_number":1,"created":"2023-03-22 13:28:18.000000000","uploader":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"ref":"refs/changes/68/878168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/68/878168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/878168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/878168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/68/878168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/68/878168/1"}}},"commit":{"parents":[{"commit":"042afb2e9aa16011b5e60c00d0fbf25caf76db8d","subject":"Merge \"[stable-only] Pin tox \u003c4\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/042afb2e9aa16011b5e60c00d0fbf25caf76db8d"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-10-03 11:38:29.000000000","tz":120},"committer":{"name":"Alan Bishop","email":"abishop@redhat.com","date":"2023-03-22 13:28:18.000000000","tz":0},"subject":"RBD: Default rbd_secret_uuid to the cluster FSID","message":"RBD: Default rbd_secret_uuid to the cluster FSID\n\nCurrently the `rbd_secret_uuid` configuration option has no default\nvalue, but at the same time the RBD driver doesn\u0027t complain if it\u0027s\nmissing, and will only fail on the Nova side if there is no\n`rbd_secret_uuid` configured in the `[LIBVIRT]` section of `nova.conf`.\n\nUsing Cinder\u0027s `rbd_secret_uuid` has been the preferred way since the\nOcata release.\n\nMost deployments set the `rbd_secret_uuid` value in libvirt\u0027s secret to\nthe Ceph cluster FSID value, and that\u0027s what this patch does when it is\nnot defined.\n\nIt doesn\u0027t change how replication locations are configured, and those\nwill still require the secret uuid to be defined.\n\nChange-Id: I739ae6ae5b4d9b074d610f6a70371b294a3e70f8\n(cherry picked from commit 6464d37d1661f5dfd58edec399a8e7e39c757b03)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/30b70ac5a07daf8feb82c8bc51a579b0fd0b591f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/30b70ac5a07daf8feb82c8bc51a579b0fd0b591f"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
