)]}'
{"id":"openstack%2Fcinder~878918","triplet_id":"openstack%2Fcinder~master~I391e03426c42b7af122722cd851c4aa03f2dcb68","project":"openstack/cinder","branch":"master","topic":"pylint-20230329","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2023-04-12 16:48:53.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-04-12 15:26:31.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}}},"hashtags":[],"change_id":"I391e03426c42b7af122722cd851c4aa03f2dcb68","subject":"Shut up pylint about win32_disk_size","status":"MERGED","created":"2023-03-29 15:14:55.000000000","updated":"2023-04-23 00:02:43.000000000","submitted":"2023-04-12 16:48:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878918-pylint-20230329","meta_rev_id":"d9002031463336510aa919b9d7be2079685db04b","_number":878918,"virtual_id_number":878918,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-12 16:48:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2023-04-12 12:25:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2023-04-12 15:26:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2023-04-12 15:26:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}],"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-29 15:36:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-29 15:39:05.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-29 16:29:38.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-03-29 16:30:55.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-29 16:39:19.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-29 17:07:57.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-29 17:14:51.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-03-29 17:35:46.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-03-29 17:38:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-29 17:54:40.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-03-29 18:13:51.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-03-29 18:43:17.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-29 20:50:24.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-30 13:24:59.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-03-30 16:32:58.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2023-03-30 16:40:26.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-03-30 18:02:01.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-03-30 20:30:28.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-31 04:53:53.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-03-31 06:31:50.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-03-31 14:31:14.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-04-01 13:03:03.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-11 17:35:51.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2023-04-12 01:05:04.000000000","updated_by":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"reviewer":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"state":"CC"},{"updated":"2023-04-12 10:07:20.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-12 12:25:57.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-04-12 15:26:31.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"}],"messages":[{"id":"fab1bb248e3103392d159b79c35515aecda467fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-03-29 15:14:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19890d8e520414359f5363bd3cfc4fc9dbd4310c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 15:36:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87a6f3fa47a34c269538eb86995c0548\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b67c20a211e04aee95a493cda04e1211 : SUCCESS in 19m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3a6121457cd0492e8ae706116d4d52c0 : SUCCESS in 8m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b5912499cabb333038fad9b83761967b6d22dc5","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-29 15:39:05.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder finger://openstack-ci-fc/bf8af5b0271040f0ae0d3ae3f1ff7233 : RETRY_LIMIT in 17m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7bc7b3fcf3b7caa795f9212dcb630cc5d2bc8fe4","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-29 16:29:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/18/878918/1/check/iscsi-driver-alletra-6k/e52e6f9/e52e6f9.html : FAILURE in 1h 13m 32s","accounts_in_message":[],"_revision_number":1},{"id":"895d4dff436d83fc4ce543d4669ef4c222ced40c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-29 16:30:55.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/54849196057145ae819e2d7a3050628b : SUCCESS in 1h 14m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ac818a5aadef3976063c8f0b20e52e9b4d63fcea","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-03-29 16:39:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/18/878918/1/ : POST_FAILURE in 1h 16m 28s","accounts_in_message":[],"_revision_number":1},{"id":"cd455586a23a463e3be9a2e26a03ff9e59cecbcf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-29 16:48:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/878918 : SUCCESS in 1h 31m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0cce56c88c88ef9b72150384c87c9fbbd497c81a","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-03-29 17:07:57.000000000","message":"Patch Set 1:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/18/878918/1/check/fujitsu-eternus-dx-iscsi/8cda429 : SUCCESS in 1h 52m 25s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/18/878918/1/check/fujitsu-eternus-dx-fc/dc8ae09 : SUCCESS in 1h 41m 04s","accounts_in_message":[],"_revision_number":1},{"id":"88a95ed8261a4a84a13ecfbc625c35c9ddaa7a3b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-29 17:14:51.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/18/878918/1/check/DellEMC_Unity_FC/fb28918/DellEMC_Unity_FC/2599 : SUCCESS in 1h 32m 39s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/18/878918/1/check/DellEMC_Unity_iSCSI/c57392e/DellEMC_Unity_iSCSI/2446 : SUCCESS in 1h 59m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"e63e74e9d83dd7041bb0c62cc503a140f9145345","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-29 17:35:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/cfa3deef0cb7487a884de96dddf52dcb : SUCCESS in 1h 29m 40s","accounts_in_message":[],"_revision_number":1},{"id":"ecc4067fdaaa18f240a09d2adcbdbc46ac3ad675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 17:38:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7af2c14042f4402a9405d491e3d7f700\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/553d4ef866a845d59a456356130185d4 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbc39c63050e41248e846a51e8339619 : SUCCESS in 10m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e005eb4834c0404eaf290eba2408c0bb : SUCCESS in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/138b413cf1604ec6b5ecc9b918f8f5dc : SUCCESS in 13m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/58f17f559dde455dbec1e25a658b4788 : SUCCESS in 57m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/316877e1c17e4e31bcbefb06d7e529c2 : SUCCESS in 1h 08m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78995c02d159458a9a5e7aaf5b1ead6e : SUCCESS in 54m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/34cda3e557b645f6940d8110af66854c : SUCCESS in 17m 55s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9770dbe8b8504cdc9952e4ca25f5c134 : FAILURE in 5m 29s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/65bb74d113d9478bbdd734d03d6c8ef6 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5b0112f3b3f748318021966031d56ddc : SUCCESS in 5m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8ad778fc9ee34aab9df25d0a977bcd09 : SUCCESS in 5m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f3c7218cc4234b94ba8314241db463d2 : SUCCESS in 53m 51s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0e011e6a043a44858b836dc16e3621dd : SUCCESS in 5m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ac56dc010a55460e8d7c6cd4d7f0d2ed : FAILURE in 1h 33m 55s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/98512c7e454b4072a2b4734bebe16e43 : FAILURE in 1h 28m 28s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2121fdfa223f432396dabb68240d708d : SUCCESS in 1h 53m 30s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/baec9b201ceb4dfe85a64d6fb53ddbb9 : SUCCESS in 1h 15m 38s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/298fb1288f3c4b6a8dc56fd70dd73f34 : SUCCESS in 1h 33m 01s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e277d9f579fd4a3bae7a73cc0fbf8add : SUCCESS in 26m 03s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/5b4212bc1d7947f8bd5972cee0e6b5b3 : SUCCESS in 1h 01m 51s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f6abf9081861447c8f761906caec8965 : SUCCESS in 2h 17m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2817989a77e549819d16b3a05e92fae5 : SUCCESS in 57m 36s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dfcdf1720d7448fa0083ff024d90f69 : SUCCESS in 52m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b7b343d8e018dbe1dd6fd17183880e89f145cea1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-29 17:54:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/18/878918/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/2d3a9a3/ : SUCCESS in 1h 45m 20s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/18/878918/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/d48ee53/ : FAILURE in 2h 37m 57s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/18/878918/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/8a401ab/ : SUCCESS in 1h 48m 25s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/18/878918/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/a955437/ : FAILURE in 29m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5af13fa80a817ad05db074e0f18df47b36d40b45","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-03-29 18:13:51.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-878918-1/FC/ : SUCCESS in 2h 15m 48s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-878918-1/iSCSI/ : SUCCESS in 2h 5m 46s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"7c2038a817c090dd297846aba86057bb283ffd55","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-29 18:43:17.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-878918-1/ns5_nfs : SUCCESS in 3h 24m 44s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-878918-1/ns5_iscsi : SUCCESS in 3h 27m 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"ab6551b74aeb4d65d911f9ef8cc8022afe958ad3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-29 20:50:24.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/18/878918/1/check/tempest-dsvm-full-focal-py3/843e348/ : SUCCESS in 1h 37m 25s\n- devstack-plugin-ceph-dsvm-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/18/878918/1/check/devstack-plugin-ceph-dsvm-focal-py3/d3e99af/ : SUCCESS in 2h 07m 37s","accounts_in_message":[],"_revision_number":1},{"id":"05ca2aa0e2cd761fc7267b8eaa34cfeaecdeb3f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-03-30 13:23:36.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"00565030a30fcd2fd433f11f2a36a084bbcf68b8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-03-30 13:24:59.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job dsvm-tempest-cisco-zonemanager-job : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"4a20e5be6b72e1dd3b853e224ad0e414ecfc80b2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-30 13:39:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-primera-cinder iscsi-driver-primera-cinder : RETRY_LIMIT in 0s","accounts_in_message":[],"_revision_number":2},{"id":"f28ba34456ff60bb0aa3984106c5112010a62276","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 13:49:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8fe4c0fbefd4adbae98cf77ab16350b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f2ab6ff1bdd4611a51267fc60b09d34 : SUCCESS in 22m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4ea3397b577e48eea11e605c947b7f93 : SUCCESS in 8m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c585092f955b8ec39cc371f2077d4796ff910120","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-30 14:42:25.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/847466a6c352437297a7ef957a7ecea0 : SUCCESS in 1h 17m 02s","accounts_in_message":[],"_revision_number":2},{"id":"84c8e56866b3f6ac872a1ab209a991e91f5450e2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-03-30 14:53:48.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/878918 : FAILURE in 1h 28m 38s","accounts_in_message":[],"_revision_number":2},{"id":"69554eb6451d12896d4aea35dedb021aff847b86","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-03-30 15:10:57.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/18/878918/2/check/iscsi-driver-alletra-6k/6aa806d/6aa806d.html : FAILURE in 1h 43m 23s","accounts_in_message":[],"_revision_number":2},{"id":"8446cc40e79978db02ee8166789112f2bc7a0846","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-30 16:17:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/18/878918/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/bf3832d/ : SUCCESS in 1h 46m 19s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/18/878918/2/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/644a4f1/ : FAILURE in 2h 52m 52s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/18/878918/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/7681ff9/ : SUCCESS in 1h 44m 00s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/18/878918/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/402ca56/ : FAILURE in 31m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c95798dbdd1ccb58a13771c4bfc1b226536cd31d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 16:31:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bdc99017c9d4c73b8c8a6f1ffdf43d7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/343dca3dd7cf4158bbf358746a56d4e5 : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab08247d2d984585b492b165bdb108b6 : SUCCESS in 11m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da81eb94526640e5b6f549d0057085c9 : SUCCESS in 12m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e58264ae25c4d84b393c28ad87fde2a : SUCCESS in 12m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/980a5fa1258647cbbaa198629a5d467f : SUCCESS in 1h 02m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/25db840348bb4f1ebd30ae45fa929a23 : SUCCESS in 1h 30m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5db38b719e10411fa9ac960d763ffb62 : SUCCESS in 59m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d6bc0608fe0a445f97d0b7c7f575bdbd : SUCCESS in 14m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/ab725b3210ad456c9140321403b2e304 : SUCCESS in 4m 22s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c918f9983ed248f2a51bf88c359911af : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9a03e80103ca46699b3369fa4aba397c : SUCCESS in 4m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/df28b27b32a6420a9f52f6d89ee96e9c : SUCCESS in 6m 03s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/941c5eb073b648d7af240b3a68ed18b1 : SUCCESS in 48m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6db762df7a384d0f8a8424a71a94102c : SUCCESS in 6m 07s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f602c740ebbc434e8a264d9f01661346 : SUCCESS in 1h 22m 39s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/7c4ea535fdf549cf82085e11708d3a60 : TIMED_OUT in 3h 05m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/eb3c5cb9d76348a488913d4a1009e694 : SUCCESS in 1h 48m 51s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/27fc4050d19c415a89730be4b3e77c98 : SUCCESS in 1h 17m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c653faada24240f4a166edbf2edc3e7e : SUCCESS in 1h 26m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/f7f44235ba614839a2fba95b0fa9ded2 : SUCCESS in 43m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/208fd32ef62643d2bac730f65dfbdbd8 : SUCCESS in 1h 09m 45s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9703c7fc34f247f7b049eadc9caa9392 : SUCCESS in 2h 17m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd3ae9b5498e42748785ddc03a892a8c : SUCCESS in 57m 16s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/009532abf255478695a6c0ad3bf91e74 : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":2},{"id":"bdd59ee727575f835163dac11cb139873c3be7bf","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-03-30 16:32:58.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"a61ef8c36e98defcc3506c5d15e4541d71e685be","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-03-30 16:33:34.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/93dd2dd64fcf4c28969d1439ac7021ef : SUCCESS in 1h 12m 58s","accounts_in_message":[],"_revision_number":2},{"id":"258637970b17b0779d37dcf7a3ed3e02745f24f9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-03-30 16:40:26.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/30_03_2023_16_31_47\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"ef4ceec2262cf2f612ebe50e9f16c82ccbbeffb4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-30 16:52:57.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-878918-2/ns5_nfs : SUCCESS in 3h 26m 2s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-878918-2/ns5_iscsi : SUCCESS in 3h 26m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"6d1a34d3530b68de97564c657a23f46ff74f291c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-30 17:07:48.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/18/878918/2/check/DellEMC_Unity_FC/734cd4b/DellEMC_Unity_FC/2604 : SUCCESS in 1h 38m 44s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/18/878918/2/check/DellEMC_Unity_iSCSI/ff2f10e/DellEMC_Unity_iSCSI/2451 : SUCCESS in 1h 54m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"afcf5cca8beb593956429396465db0ec93bafabd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-03-30 18:02:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/878918/2/upstream-check/cinder-tempest-plugin-ontap-fcp/675424e/ : SUCCESS in 1h 29m 35s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/878918/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/ad25e74/ : SUCCESS in 44m 26s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/878918/2/upstream-check/cinder-tempest-plugin-ontap-nfs/3b249c9/ : SUCCESS in 42m 45s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/878918/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/ea6fe97/ : SUCCESS in 1h 22m 11s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"67b51b5c4e12bfa985f31876e1021cdda03e7c59","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-03-30 18:11:35.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/18/878918/2/check/tempest-dsvm-full-focal-py3/7a7d95f/ : SUCCESS in 1h 40m 50s\n- devstack-plugin-ceph-dsvm-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/18/878918/2/check/devstack-plugin-ceph-dsvm-focal-py3/d1a0cb8/ : FAILURE in 2h 57m 00s","accounts_in_message":[],"_revision_number":2},{"id":"0bc84f4eeb2626c3c26399dc3052adb3022a7982","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-03-30 18:40:12.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-878918-2/FC/ : SUCCESS in 2h 33m 10s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-878918-2/iSCSI/ : SUCCESS in 2h 20m 28s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"88b615bf555a334b96506ad166889457542ae77e","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-03-30 18:45:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e979ee50cafeb943042cd2e51ae5df4a8fda2d80","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-03-30 20:30:28.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/18/878918/2/check/DellEMC_PowerFlex/f2a34bf/DellEMC_PowerFlex/1799 : SUCCESS in 1h 28m 48s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"68bbb75c51178a59459f892cadef25cf79851e0e","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-03-30 21:16:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/18/878918/2/ : POST_FAILURE in 1h 18m 19s","accounts_in_message":[],"_revision_number":2},{"id":"4434a15f2d59d1f199c05f95e2a7f5b05758b706","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-03-31 04:53:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/878918-2/HPMSA-iSCSI/ : SUCCESS in 1:37:34\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/878918-2/HPMSA-FC/ : SUCCESS in 1:16:28","accounts_in_message":[],"_revision_number":2},{"id":"f39d660a222657b80e37a19fab14439894f84f61","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-03-31 06:31:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/878918-2/PVME-iSCSI/ : SUCCESS in 1:46:20\n- PVME-FC https://os-logs.seagate.com/dell-ci/878918-2/PVME-FC/ : SUCCESS in 1:28:54","accounts_in_message":[],"_revision_number":2},{"id":"dfa17e1f1135a06346f0dbc61ad68f75e7441d03","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-03-31 14:31:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/878918/2/20230331174642/NEC-V-ISCSI : SUCCESS in 5h 25m 54s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/878918/2/20230331174642/NEC-V-FC : SUCCESS in 5h 44m 29s","accounts_in_message":[],"_revision_number":2},{"id":"af447c154a7a96b75d8e4ae67baa3f5894ba80e3","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-04-01 13:03:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/18/878918/2/20230401180226/NEC-FC : SUCCESS in 4h 00m 19s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/18/878918/2/20230401180226/NEC-ISCSI : SUCCESS in 4h 00m 33s","accounts_in_message":[],"_revision_number":2},{"id":"7b8a32e8c25c362d6e06316505cc8ad7a1e0206e","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-04-11 17:35:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"06372a4c919eafb4cf9b416533121af87d1b6551","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-04-11 18:02:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a08f29abc75fc4487277ad318a98387f7630ae2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-04-11 18:18:58.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c63a9769f8f961f8b164eeb194bb9670d3cd49ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 18:30:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e15519b192fc4ea7a29edc487d191bcd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05d717fd1e114d359302a03b26107491 : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/331e104f3f0b4c5b9946a7297c40d35c : SUCCESS in 9m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"63740809bab168ffbd7219d058dcf9f286e67a46","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-04-11 19:22:11.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a0377558c7c443cf9f853b2b863cdcbc : SUCCESS in 1h 01m 41s","accounts_in_message":[],"_revision_number":3},{"id":"93cdc2e607b6f8851cf46e37e83dd9949c4ee9dd","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-04-11 19:40:07.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/5be1d8df37c14b6795e764b676d284ec : SUCCESS in 1h 20m 22s","accounts_in_message":[],"_revision_number":3},{"id":"8428081d945e5db54c8ae7181db52c7d45eb5cf7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-04-11 19:53:35.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/878918 : FAILURE in 1h 32m 09s","accounts_in_message":[],"_revision_number":3},{"id":"9914f28947669823539ed4e6fdb1ec8582af0bcc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-11 20:19:19.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/18/878918/3/check/iscsi-driver-alletra-6k/3df0e2f/3df0e2f.html : FAILURE in 1h 58m 04s","accounts_in_message":[],"_revision_number":3},{"id":"23c4d5808ee631712e04c85d0b43298a4a614188","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-04-11 20:41:53.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-878918-3/FC/ : SUCCESS in 2h 22m 32s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-18-878918-3/iSCSI/ : SUCCESS in 2h 18m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"c4b4701a7bfcf8dcb83a3e83902816879e1a6fc8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-04-11 20:44:25.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/18/878918/3/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/3c04537/ : SUCCESS in 1h 43m 48s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/18/878918/3/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/28772d3/ : FAILURE in 2h 08m 00s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/18/878918/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/451e0d3/ : SUCCESS in 1h 49m 33s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/18/878918/3/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/a39f2f8/ : FAILURE in 34m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"40b706e2c767de45f32d3f3619c8d68254199c0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 21:07:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1d221d3fd4345d791d98cf8627c8421\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f432841b840b45459f7442348d3b3a94 : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/907676677200455daf9e826d8e0dea48 : SUCCESS in 11m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ccd55f252cf14cec8635d370749f2519 : SUCCESS in 10m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38f2e394bb0f4c95a12f5ae746a05adf : SUCCESS in 12m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/bf496932d5874fb093c9942444b02899 : SUCCESS in 1h 02m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a8a4d2c8d4064c708754fb5173812774 : SUCCESS in 1h 21m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e527a09c49db48449098d7fa9af02309 : SUCCESS in 1h 02m 54s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f925fe9411294e179361addfaddcf6ee : SUCCESS in 19m 09s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/af019968c6aa43f193ae04cb3babf4b3 : SUCCESS in 4m 31s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/be2e9a4709584bcfbe31ded594d103a5 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5836689d51964db0bbd256cabe938262 : SUCCESS in 5m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/19ad6713cccb4f07b704e4c01ea0b07f : SUCCESS in 7m 31s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/1ff046d6168f4b858ee7f8b9ddcf1008 : SUCCESS in 32m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/def44c9acd414fdaabd3d9138028ebbe : SUCCESS in 6m 26s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/58ef8a4583b546fda361515ef078e919 : SUCCESS in 53m 22s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/387b907397484ed8b41661abe73923a9 : FAILURE in 1h 06m 40s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d13bee79bac548baa2701b5401b34c2b : SUCCESS in 2h 13m 02s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/9b2b35dfc019411e9d3842f6692c2526 : SUCCESS in 1h 12m 17s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e26a385b74a147eea72944054b08157d : SUCCESS in 1h 25m 24s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/cff09a899f00441f8dcf1ed8abc9e3a5 : SUCCESS in 45m 32s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/766fc14bcb6a4aa09666442ddf3bdaaa : SUCCESS in 1h 13m 25s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/70b1c3f0e5fb439db471c7d4e74142ac : SUCCESS in 1h 52m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33ce149735ee4c62b667e2e5b59f797b : SUCCESS in 53m 26s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98cfbc8ea6d946c581c2c3f57f3820a3 : SUCCESS in 1h 51m 08s","accounts_in_message":[],"_revision_number":3},{"id":"d7488658dbdda5495839cbc59730e8f2f740b72d","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-04-11 21:09:19.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"ebe026482b34663759e0f3433ecd4eb686b46bc6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-04-11 21:48:33.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-18-878918-3/ns5_nfs : SUCCESS in 3h 22m 35s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-18-878918-3/ns5_iscsi : SUCCESS in 3h 26m 22s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"8f12c299f8a36151be26720640f83fb2a722b110","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-04-11 22:28:26.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/878918/3/upstream-check/cinder-tempest-plugin-ontap-fcp/8062a24/ : SUCCESS in 1h 00m 09s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/878918/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/c50949e/ : SUCCESS in 1h 16m 14s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/878918/3/upstream-check/cinder-tempest-plugin-ontap-nfs/533193a/ : SUCCESS in 1h 20m 13s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/18/878918/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/a5ae4b0/ : FAILURE in 1h 15m 35s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"f4c2123c20e8625ca17806038188619a575b3782","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-04-11 22:36:06.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/18/878918/3/check/DellEMC_PowerFlex/a794eaf/DellEMC_PowerFlex/1831 : SUCCESS in 1h 28m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"e22518f61294bc8c3df7aeea939b5af588f21146","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-04-11 23:58:50.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/18/878918/3/ : FAILURE in 1h 16m 58s","accounts_in_message":[],"_revision_number":3},{"id":"1a15b88e538a0e00147bae07cf73316faf8e93d7","author":{"_account_id":34162,"name":"HPE XP Storage CI","display_name":"HPE XP Storage CI","email":"hpexpci@hpe.com","username":"hpexpci"},"date":"2023-04-12 01:05:04.000000000","message":"Patch Set 3:\n\n* hpe_xp_fc http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/18/878918/3/fc : SUCCESS in 2h 07m 24s\n* hpe_xp_iscsi http://ec2-13-56-246-116.us-west-1.compute.amazonaws.com/refs/changes/18/878918/3/iscsi : FAILURE in 3h 03m 53s","accounts_in_message":[],"_revision_number":3},{"id":"aea67173276d6f339f92ff79a0b04f10cf3d8b4e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-04-12 03:12:40.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/18/878918/3/check/dsvm-tempest-cisco-zonemanager-job/c35a0cc/ : SUCCESS in 1h 04m 55s","accounts_in_message":[],"_revision_number":3},{"id":"e341792a9e7ae52782917c453e207b9ae5ecb1bf","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-04-12 03:31:52.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/878918-3/Seagate-iSCSI/ : SUCCESS in 1:46:38\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/878918-3/Seagate-FC/ : SUCCESS in 1:29:29","accounts_in_message":[],"_revision_number":3},{"id":"66232fcea71d00cb89a7fe4867aaad5dffb62fa1","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-04-12 03:37:29.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/878918-3/PVME-iSCSI/ : SUCCESS in 1:51:49\n- PVME-FC https://os-logs.seagate.com/dell-ci/878918-3/PVME-FC/ : SUCCESS in 1:31:38","accounts_in_message":[],"_revision_number":3},{"id":"748a61c950a64eec1e0d39b90cfa34bcf5160adf","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-04-12 08:26:49.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/18/878918/3/check/tempest-dsvm-full-focal-py3/4f1b405/ : SUCCESS in 1h 53m 24s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/33c68fd332fd443982e007fdf41219c4 : RETRY_LIMIT in 1m 02s","accounts_in_message":[],"_revision_number":3},{"id":"d91b2591441aac125336d41f44a00481ed16a7e4","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-04-12 10:07:20.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/18/878918/3/check/DellEMC_PowerMAX_FC/687f593/DellEMC_PowerMAX_FC/865 : ABORTED in 43s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/18/878918/3/check/DellEMC_PowerMAX_iSCSI/86d7f88/DellEMC_PowerMAX_iSCSI/847 : ABORTED in 14s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"a50ec0ad50a6136b1b3d6cc15d0e8c99e0f9a8d9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-04-12 12:25:57.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"894c75071d0697567c2fccf052ea1017d39a0b35","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-04-12 15:26:31.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b9cb28367708e2b99d035b86d481d655847a126c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 15:27:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b555edb97bec43d7f06e35b697468e6459f030cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 16:48:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1da8aed4bdf48b99cb487517a2a1cd4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21d34d1e2c314cdf90c217a0c018aac4 : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5c97a78a7a24452baa90da95973321d : SUCCESS in 10m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24345c8ac6d242ca90895e4d7961ea4f : SUCCESS in 12m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c50ecac2c6564a6c8834d875a284827d : SUCCESS in 13m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/e4fc1186618d4cf59c2faaa0f2b22f56 : SUCCESS in 49m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/69792840f137416fbf1b20693d62c062 : SUCCESS in 1h 12m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/226c663d99de4d4dbc0d11c7f11ef87d : SUCCESS in 55m 16s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e1d9aae3788342808eb592faf601abf8 : SUCCESS in 1h 15m 52s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/daf6ea6f83ca4b90b9dd2c313ea3f16e : SUCCESS in 1h 05m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0fec1713a7974c478e1f87275fe89c3f : SUCCESS in 55m 00s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81ac6d93a23b4d8c8e9a99c80724e998 : SUCCESS in 1h 10m 39s","accounts_in_message":[],"_revision_number":3},{"id":"63f8d382c2bf26b2e1a7ca270ddc33e11fac393f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 16:48:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"afd70b0f7d36f9e08e7fe9c91900a3429026bc5f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 16:50:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a51047387b2d4f7986b29200eac929a1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57e1fdd0f6ee49cf9850faf8ae01a116 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":3},{"id":"d9002031463336510aa919b9d7be2079685db04b","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-23 00:02:43.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/18/878918/3/check/iscsi-driver-nimble/835dcdf/835dcdf.html : FAILURE in 33m 34s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"63b6e569602a05801b0706a646f753af9b2bf0da","revisions":{"1241c829a7cc5e64a61f7f79e2630a21f468ee85":{"kind":"REWORK","_number":1,"created":"2023-03-29 15:14:55.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/18/878918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/878918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/878918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/878918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/878918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/878918/1"}}},"commit":{"parents":[{"commit":"3fc37548042fcdc745d35a0a902213482fc4422c","subject":"Merge \"Tests: Remove typo\u0027d assert calls from test_attachments_manager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/3fc37548042fcdc745d35a0a902213482fc4422c"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2023-03-29 15:13:06.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2023-03-29 15:13:06.000000000","tz":-300},"subject":"Shut up pylint about win32_disk_size","message":"Shut up pylint about win32_disk_size\n\nThe pylint job is non-voting, but it was annoying to see it fail.\n\nTo add the insult to injury, the code was correct, but the pylint\nwas not smart enough. So this patch only exists to address\nshortcomings of our toolchain.\n\nChange-Id: I391e03426c42b7af122722cd851c4aa03f2dcb68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1241c829a7cc5e64a61f7f79e2630a21f468ee85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/1241c829a7cc5e64a61f7f79e2630a21f468ee85"}]},"branch":"refs/heads/master"},"d025aae5f513982d4721ffc4fb4633b23cfcc4c9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-30 13:23:36.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/18/878918/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/878918/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/878918/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/878918/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/878918/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/878918/2"}}},"commit":{"parents":[{"commit":"19250d3f24e64b28e1d079a81759d533e5c22753","subject":"Merge \"Bump mypy to 1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/19250d3f24e64b28e1d079a81759d533e5c22753"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2023-03-29 15:13:06.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2023-03-30 13:23:29.000000000","tz":-300},"subject":"Shut up pylint about win32_disk_size","message":"Shut up pylint about win32_disk_size\n\nThe pylint job is non-voting, but it was annoying to see it fail.\n\nTo add the insult to injury, the code was correct, but the pylint\nwas not smart enough. So this patch only exists to address\nshortcomings of our toolchain.\n\nChange-Id: I391e03426c42b7af122722cd851c4aa03f2dcb68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d025aae5f513982d4721ffc4fb4633b23cfcc4c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d025aae5f513982d4721ffc4fb4633b23cfcc4c9"}]},"branch":"refs/heads/master"},"63b6e569602a05801b0706a646f753af9b2bf0da":{"kind":"REWORK","_number":3,"created":"2023-04-11 18:18:58.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/18/878918/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/18/878918/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/878918/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/878918/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/18/878918/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/18/878918/3"}}},"commit":{"parents":[{"commit":"dfef52f3a6e6ba6fde0e985621673064c8c86356","subject":"Fix wrong assertion methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/dfef52f3a6e6ba6fde0e985621673064c8c86356"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2023-03-29 15:13:06.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2023-04-11 18:02:26.000000000","tz":-300},"subject":"Shut up pylint about win32_disk_size","message":"Shut up pylint about win32_disk_size\n\nThe pylint job is non-voting, but it was annoying to see it fail.\n\nTo add the insult to injury, the code was correct, but the pylint\nwas not smart enough. So this patch only exists to address\nshortcomings of our toolchain.\n\nChange-Id: I391e03426c42b7af122722cd851c4aa03f2dcb68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63b6e569602a05801b0706a646f753af9b2bf0da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/63b6e569602a05801b0706a646f753af9b2bf0da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
