)]}'
{"id":"openstack%2Fcinder~880492","triplet_id":"openstack%2Fcinder~stable%2Fzed~I0d266a57f68221a3b1740a7376e152bb64cac729","project":"openstack/cinder","branch":"stable/zed","topic":"bug/1990136-stable/zed","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-04-29 11:20:37.000000000","reason":"Change was abandoned"},"31779":{"account":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"last_update":"2024-04-29 11:20:37.000000000","reason":"Change was abandoned"},"31016":{"account":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"last_update":"2024-04-29 11:20:37.000000000","reason":"Change was abandoned"},"7198":{"account":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"last_update":"2024-04-29 11:20:37.000000000","reason":"Change was abandoned"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-04-29 11:20:37.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0d266a57f68221a3b1740a7376e152bb64cac729","subject":"Dell PowerFlex: Additionnal params for enabling self signed certificates","status":"ABANDONED","created":"2023-04-14 12:46:57.000000000","updated":"2024-04-29 11:20:37.000000000","total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":858370,"cherry_pick_of_patch_set":5,"meta_rev_id":"9b460d1b9e40e7de476cd36f7859b9163a0cd53a","_number":880492,"virtual_id_number":880492,"owner":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-03-28 09:40:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-14 12:58:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-14 13:06:43.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-14 13:14:02.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-04-14 13:46:46.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-14 13:48:54.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-14 14:16:30.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-04-14 15:28:54.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-04-14 15:55:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-15 09:11:48.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-04-16 15:07:01.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-17 13:41:11.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-04-17 13:41:11.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-04-19 14:16:03.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2023-09-18 07:11:07.000000000","updated_by":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"reviewer":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"state":"REVIEWER"},{"updated":"2024-04-29 11:20:37.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"dd3fb21d0f3790121df2222301badba728bdf517","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-04-14 12:46:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"20dd3c37bda075cb3dc84b4a611d1c60c4a8fbd8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 12:58:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c970a1a449b94ea3b53242b1ced8eaf0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4029355a3345486391546bd70fae6bca : SUCCESS in 9m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/756beabcff524b43a7276a8d0e470f34 : SUCCESS in 9m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0982d8e3a7701939d4ed5f3d29df9e66cbe05181","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-04-14 13:06:43.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-92-880492-1/ns5_nfs : FAILURE in 1m 30s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-92-880492-1/ns5_iscsi : FAILURE in 1m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"bef65977d6c7283595fc90da00716e1a95a9b10a","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-04-14 13:14:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/92/880492/1/check/dsvm-tempest-cisco-zonemanager-job/6b6e2fa/ : FAILURE in 24m 41s","accounts_in_message":[],"_revision_number":1},{"id":"aad9d3cff6ba767620910f38dbaf68220487174e","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-04-14 13:46:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/92/880492/1/ : FAILURE in 56m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7c09fdf39a34b67fddca6c4ba3d86dc9aafc13cf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-04-14 13:48:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/880492 : FAILURE in 59m 54s","accounts_in_message":[],"_revision_number":1},{"id":"ad8ed37da327d87faab5454a8576764f99079738","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-14 14:16:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/92/880492/1/check/iscsi-driver-alletra-6k/330eeb4/330eeb4.html : FAILURE in 1h 28m 26s","accounts_in_message":[],"_revision_number":1},{"id":"899ebb08bf6e6f40c4755bf8386a0d2ecb177655","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-04-14 15:28:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio http://openstack-logs.purestorage.com/92/880492/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio/ea8d139/ : SUCCESS in 1h 39m 46s\n- pure-devstack-cinder-tempest-fc-aio http://openstack-logs.purestorage.com/92/880492/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio/8f0af64/ : FAILURE in 2h 39m 34s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio http://openstack-logs.purestorage.com/92/880492/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio/5b2efcd/ : SUCCESS in 1h 44m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d1a1941fb4b9b8e8d0a73a16c0525ddbba5f2c32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 15:55:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7877fe5e039b4bbab50ca8796d0f5803\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a0d0c7379b64bd1898100d3bceb110b : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fe285ef46d747e3a51f6e58dce21ff1 : SUCCESS in 10m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1542b4455814741b5a5751c0a2c5956 : SUCCESS in 8m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e71bbd1d95d445ba6843a6685de8e79 : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f8fddb3121748799b5e866024e5984d : SUCCESS in 11m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/18847732733f48d088dff1279c107a83 : SUCCESS in 44m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a0ac0771b04e46e28058aed032b5add2 : SUCCESS in 1h 15m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca8965a542564f888a3b2b84910cec11 : SUCCESS in 53m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bd6ebe286874345b7c8809486ec7fc8 : SUCCESS in 7m 23s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/c72b7dc61e78469b95986a01821ef669 : SUCCESS in 13m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/a3baabff0062431d986b83bea4415acf : SUCCESS in 4m 18s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5c67faf2282540c49b48614f25268293 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4c7fba7ac08246cd90a3e24bc6d9cee9 : SUCCESS in 5m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b78ec6474bcf4092954cad7702e4d148 : SUCCESS in 5m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8fd518e4ecd84042b27f96437da7edf0 : SUCCESS in 30m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/93b7579edf7645ca8f9421465555e5be : FAILURE in 6m 29s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c3548815c07445f587684a479ffc6e7e : SUCCESS in 1h 00m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/601ac99550374a608522e674179f5476 : TIMED_OUT in 3h 04m 20s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/55332e562af4451a937c576c7f54af67 : SUCCESS in 1h 24m 03s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/cb427e2a37d2490dbf2df073241b0ca3 : SUCCESS in 1h 04m 29s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/c49cd613f73841d8bf4a1c701b138b15 : SUCCESS in 1h 16m 57s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/5bd2e1c1afae4e4bbd659f8a02e3e493 : SUCCESS in 32m 13s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/112a086c5cb64ea79971c5d5b8fb6aa2 : SUCCESS in 1h 11m 27s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f6af5b54bcfd42e8bed891065d1b73cb : SUCCESS in 1h 55m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ce5e12c8ad44cdfbb7484c69ffbe53d : SUCCESS in 45m 43s","accounts_in_message":[],"_revision_number":1},{"id":"620f2283fcac7c59104a36c043602a713c695bee","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-04-15 09:11:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/880492/1/20230415151744/NEC-V-ISCSI : SUCCESS in 2h 33m 51s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/880492/1/20230415151744/NEC-V-FC : SUCCESS in 2h 54m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a025ee3699500df4db2f27c2ac6848b6b71efa70","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-04-16 15:07:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/880492/1/20230416171225/NEC-FC : SUCCESS in 3h 32m 57s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/880492/1/20230416171225/NEC-ISCSI : SUCCESS in 6h 54m 34s","accounts_in_message":[],"_revision_number":1},{"id":"0a2f41311d7077ba6652b11b384934c47dc99fc7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-18 08:50:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/92/880492/1/check/iscsi-driver-nimble/7b42d4f/7b42d4f.html : FAILURE in 52m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7396ef30d8a68a2927493e6b94a0dacbc8ffa8b2","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2023-04-19 14:16:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83d7f8bbafc456d487dfdc7a2085d6338d7ee6b7","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-09-14 06:56:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0cd642642cf661aafaad20e7ea3c42ef3612731","author":{"_account_id":31016,"name":"Ivan Pchelintsev","email":"i.pchelintsev@yadro.com","username":"pcheli"},"date":"2023-09-18 07:11:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0badaff333901e57806a0cedc56d66513bece6a9","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-09-20 07:58:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3b919721a701182d872735fa97271d55e37e0bdc","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2023-11-08 15:50:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0aed0279aa4c80be09303c5f734fca0265a5f128","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-09 15:48:26.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"49244145006dc7cd5328a2ad5bd13fdaffaa0a1d","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2024-02-19 08:40:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd6bafc418c531ae8b9647c409efc0ad5024c957","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-03-26 12:30:51.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cbd735a1f78b4dbd86dd98b802368139631dd7e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2024-03-26 12:44:10.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"89a1bc39f8ae63d0bb26dcec06d72cd084073ef3","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2024-03-26 12:44:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cbe0160cc13506b5d722ea903b42018a4ede3dfc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-26 12:50:35.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/7176a05e9c8a495b90294db5d1187f7a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/3f682ab0183a44809ba34de2867af3be : NODE_FAILURE Node request 200-0000002883 failed in 0s","accounts_in_message":[],"_revision_number":2},{"id":"120455fd8a987e9c9eef59af755ed917d5cf39b5","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-26 13:00:44.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/880492 : FAILURE in 14m 59s","accounts_in_message":[],"_revision_number":2},{"id":"956613afd94d81f0d09b457c07bd65d46459b9e7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2024-03-26 13:00:49.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/880492 : FAILURE in 15m 19s","accounts_in_message":[],"_revision_number":2},{"id":"6b9a4ea21a358dc2e428db80d223028d67c47e77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 13:09:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9648c7ed0923445fbb4907a98ec7f225\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/92de342867ee46f2962b4a79b044e471 : SUCCESS in 17m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/520bb72ec3044b1ab78837274f1a6e5c : SUCCESS in 21m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3c8f60a84d235150a040f99b800b14544dd3831a","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2024-03-26 13:57:58.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://github.com/hpe-storage/nimble-ci-logs/tree/master/new_logs/880492 : FAILURE in 53m 24s","accounts_in_message":[],"_revision_number":2},{"id":"90b45f38846016c7efa156d24081a9f68599b748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-26 14:41:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffa5f9a1137b4fb5be34a9144c61129c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a46cb4c2fa214299af78da8ad2cf16a7 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d928cf503ac4db9a5b937bf34b9227a : SUCCESS in 11m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39a8c3c5a69d405f84939aaf86394d2c : SUCCESS in 11m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f6ccd43498541e796760aea3bef4841 : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/521b08f6a75f4e2c8d83913cc49994b1 : SUCCESS in 12m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c69ae2dd5864ec8b3986a0284c5adbf : SUCCESS in 14m 14s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5a9834aab7fc49d2b6a43406f7459531 : SUCCESS in 13m 29s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c52562e5284548f8b960542b0d310b10 : SUCCESS in 6m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6743d7a76a9c4f19a6f760e8a2cb7e20 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/90ddcc24ee0e4692ba9c281a4fab1e70 : SUCCESS in 5m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f2aac1e5860b45ba8383eb0b82da60dc : SUCCESS in 6m 42s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9654d9fa5fef4b02912642cfa2cca248 : SUCCESS in 28m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1055994596a94277ac54b1a78f682cc0 : FAILURE in 5m 54s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7150c6f22c974bdaa7d3e27d38189622 : SUCCESS in 1h 51m 31s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/46a92b9c408f4977ac03d97b0e24e089 : FAILURE in 1h 25m 48s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/dcd02aa416ee468d8b6266c22d608059 : SUCCESS in 1h 32m 10s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0c66b68c93804a20b6048e2e2fc1da18 : SUCCESS in 1h 07m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/9569082a79d342799a63055f859828f6 : SUCCESS in 46m 43s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1923bbd2ef8f4a75922af419aef93ff0 : SUCCESS in 1h 16m 08s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bd6402f72cd14f128cfea26f92cf6a05 : FAILURE in 1h 04m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a5bf350f7534e05b78d6a25788c90a4 : FAILURE in 1h 07m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42054bfdee0242cda567bcd1a48d1934 : SUCCESS in 57m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/938c1ee55df84959820b576fa48aa67d : SUCCESS in 39m 00s","accounts_in_message":[],"_revision_number":2},{"id":"aa2d52d0414169a7992cb53daa6fb0ba0e8a5e2c","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-03-26 23:46:14.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/880492/2/20240326214902/NEC-V-ISCSI : FAILURE in 9h 16m 58s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/880492/2/20240326214902/NEC-V-FC : FAILURE in 10h 57m 08s","accounts_in_message":[],"_revision_number":2},{"id":"aeac7d51557d3c3e9df539b399812bed5e97c576","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-03-27 05:56:37.000000000","message":"Patch Set 2:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/92/880492/2/20240327030743/NEC-ISCSI : FAILURE in 11h 19m 00s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/92/880492/2/20240327030743/NEC-FC : FAILURE in 11h 48m 51s","accounts_in_message":[],"_revision_number":2},{"id":"2957cab06c764a4028c2498e8fd6a87c3c97dba8","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2024-03-28 07:50:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bfc41b8cdae242255a31ed339ac5dc7f541f7c27","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2024-03-28 07:59:21.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/57882f72ea5e4f539a8178a481376949\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/baffc14ac74441179d53da1379d0f64b : NODE_FAILURE Node request 200-0000002890 failed in 0s","accounts_in_message":[],"_revision_number":2},{"id":"b180ed7fac619d94a7c06759f30fa0d814868308","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 08:09:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f1c37d5c5024716bbac368462dd6601\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74d2f68043b84631a20bf6512fb70c19 : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/30657082341641af95712362529c05ed : SUCCESS in 17m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3ac28b8cde4faa625a59df10b20f6431e2d33c08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 09:40:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7075db25b6e94d95bc50eea342535778\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/741bb119c7de443baacbbecebb1e502e : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48f3e9db42ce414cb518f4784d976fc8 : SUCCESS in 8m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3a8d546a00040c99a8c7b1ce3d7fc9b : SUCCESS in 8m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d326881f6c743009315770a4721dac3 : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6852652bda14442aadf3148b0d95a471 : SUCCESS in 9m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13880b2a52a943b89520789fb9312999 : SUCCESS in 13m 26s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/37667ea32d8f4b9aad8177f9c0b42fca : SUCCESS in 14m 26s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9a76124a3b5f4fb4accf791b3ae84eb9 : SUCCESS in 4m 57s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f5c72c02555b4ad19a4586070854e5d1 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0d0c367bcf1e4dda9be25eedd8ffe860 : SUCCESS in 5m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c0198f9819b945b187960a373dcef610 : SUCCESS in 6m 17s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/988051d4d208436684bb8a0877c90689 : SUCCESS in 39m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fbc9ad956f674e968e06bcc59cac730b : FAILURE in 5m 16s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2c5cec5052bb483d9a405d5c6ef3b33e : SUCCESS in 1h 41m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/82aefe8c8ae74202a78e9e07b9431c4b : FAILURE in 1h 21m 38s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5e2b740a5d174bf2ba7ec62e8c615cb7 : SUCCESS in 1h 11m 16s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/81d9821200c5469f90f4ddbb81eda0c6 : SUCCESS in 1h 24m 22s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b1e00a692a404e9499f8d65373343828 : FAILURE in 32m 57s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/2ce1b36ed758406a984a53bd3a4358b8 : SUCCESS in 59m 53s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a47e18c451984b35a5c51a1b5145d09d : SUCCESS in 1h 00m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/23f379eee8bc4d30a2f3fc8772ab6d70 : SUCCESS in 1h 13m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbf0bb57827d41a2b86f3d23da069e7a : SUCCESS in 53m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b82b11226fd4393968abc8c0de1e9fe : SUCCESS in 52m 41s","accounts_in_message":[],"_revision_number":2},{"id":"9b460d1b9e40e7de476cd36f7859b9163a0cd53a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-29 11:20:37.000000000","message":"Abandoned\n\nstable/zed branch of openstack/cinder is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/zed if you want to further work on this patch.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4d29ae5c6592cc6cbd176a34c40676044c65f802","revisions":{"9048659b4d1302efa7a4c777ebf22fb4c5bc3155":{"kind":"REWORK","_number":1,"created":"2023-04-14 12:46:57.000000000","uploader":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"ref":"refs/changes/92/880492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/880492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/880492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/880492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/880492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/880492/1"}}},"commit":{"parents":[{"commit":"91e6ff7a5552be2e9403e58a3da0373b9749760f","subject":"Merge \"RBD: Default rbd_secret_uuid to the cluster FSID\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/91e6ff7a5552be2e9403e58a3da0373b9749760f"}]}],"author":{"name":"happystacker","email":"jeanpierre.roquesalane@dell.com","date":"2022-09-19 06:55:24.000000000","tz":120},"committer":{"name":"Jean Pierre Roquesalane","email":"jeanpierre.roquesalane@dell.com","date":"2023-04-14 12:46:57.000000000","tz":0},"subject":"Dell PowerFlex: Additionnal params for enabling self signed certificates","message":"Dell PowerFlex: Additionnal params for enabling self signed certificates\n\nInitially before the change\nhttps://review.opendev.org/c/openstack/os-brick/+/810419 was merged\nto close the bug https://bugs.launchpad.net/os-brick/+bug/1929223,\nverify_cert was always set to False which can lead to security issues.\nIt has been decided through this change that this option can be set\nto True or False based upon security requirements. This change\nintroduced a regression failure as the value set to the option\nis not part of connection_properties.\n\nThis patch adds additional params during initialization so that it\ncan be carried over os-brick and get adequate REST API response.\n\nCloses-Bug: 1990136\nChange-Id: I0d266a57f68221a3b1740a7376e152bb64cac729\n(cherry picked from commit 82823ace4d714ac10427ea3c6fed320c27b56f7d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9048659b4d1302efa7a4c777ebf22fb4c5bc3155"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/9048659b4d1302efa7a4c777ebf22fb4c5bc3155"}]},"branch":"refs/heads/stable/zed"},"4d29ae5c6592cc6cbd176a34c40676044c65f802":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-03-26 12:44:10.000000000","uploader":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"ref":"refs/changes/92/880492/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/92/880492/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/880492/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/880492/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/92/880492/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/92/880492/2"}}},"commit":{"parents":[{"commit":"91e6ff7a5552be2e9403e58a3da0373b9749760f","subject":"Merge \"RBD: Default rbd_secret_uuid to the cluster FSID\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/91e6ff7a5552be2e9403e58a3da0373b9749760f"}]}],"author":{"name":"happystacker","email":"jeanpierre.roquesalane@dell.com","date":"2022-09-19 06:55:24.000000000","tz":120},"committer":{"name":"Jean Pierre Roquesalane","email":"jeanpierre.roquesalane@dell.com","date":"2024-03-26 12:44:10.000000000","tz":0},"subject":"Dell PowerFlex: Additionnal params for enabling self signed certificates","message":"Dell PowerFlex: Additionnal params for enabling self signed certificates\n\nInitially before the change\nhttps://review.opendev.org/c/openstack/os-brick/+/810419 was merged\nto close the bug https://bugs.launchpad.net/os-brick/+bug/1929223,\nverify_cert was always set to False which can lead to security issues.\nIt has been decided through this change that this option can be set\nto True or False based upon security requirements. This change\nintroduced a regression failure as the value set to the option\nis not part of connection_properties.\n\nThis patch adds additional params during initialization so that it\ncan be carried over os-brick and get adequate REST API response.\n\nCloses-Bug: 1990136\nChange-Id: I0d266a57f68221a3b1740a7376e152bb64cac729\n(cherry picked from commit 8794f84e77435a4d86d0c08980fff18a80151dc2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d29ae5c6592cc6cbd176a34c40676044c65f802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4d29ae5c6592cc6cbd176a34c40676044c65f802"}]},"branch":"refs/heads/stable/zed","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
