)]}'
{"id":"openstack%2Fcinder~880512","triplet_id":"openstack%2Fcinder~master~I41a1d83664fca194567d61c1b4eecf6a7a6a1334","project":"openstack/cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"35002":{"account":{"_account_id":35002,"deleted":true},"last_update":"2025-08-26 17:20:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I41a1d83664fca194567d61c1b4eecf6a7a6a1334","subject":"DNM - Netapp test to add upstream-check REST jobs with file filter to trigger only in Netapp driver","status":"ABANDONED","created":"2023-04-14 14:39:16.000000000","updated":"2025-08-26 17:20:13.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"726ed35f618769291a0b0f2e98f60bcff7d13c00","_number":880512,"virtual_id_number":880512,"owner":{"_account_id":35002,"deleted":true},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-14 17:08:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-14 14:55:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-14 14:58:12.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-14 15:27:28.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-04-14 15:44:45.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-14 15:52:14.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-04-14 16:06:55.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-04-14 16:14:56.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-14 16:21:45.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-14 16:38:45.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-04-14 17:08:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-14 17:11:08.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-04-14 18:02:47.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-14 18:08:10.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-14 19:18:05.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-04-15 15:27:51.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-04-17 05:22:36.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"6f85a2ab9a0cf2d764b422368382808a069f1b67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35002,"deleted":true},"date":"2023-04-14 14:39:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b106600d71737937e1811d9bdb1a0a200fc5032","author":{"_account_id":35002,"deleted":true},"date":"2023-04-14 14:41:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"501d6ce7b56cb0961ba5303851e286cefec4877e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35002,"deleted":true},"date":"2023-04-14 14:43:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"259e023b6c5c22b1269abc7043dcf12b9e950fbc","author":{"_account_id":35002,"deleted":true},"date":"2023-04-14 14:45:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0303f0930e3d1ed0c22b6f76ca5eb7f754410972","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 14:55:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49db53b11a4940afa337937bcddbcf46\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/59358cd1abd942b5837890e3d577a7c2 : SUCCESS in 10m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/35bed201ee3848a8b7123b4feed73e9d : SUCCESS in 10m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7522bd8b5aeeb16b8547b9a0813c6dd00bdeba06","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-04-14 14:58:12.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s\n- devstack-plugin-ceph-dsvm-focal-py3 devstack-plugin-ceph-dsvm-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"e9342a94acfe387fd8b548d5c103fcc1214c8ad5","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-04-14 15:27:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/f560c392194240f792822015b7680a26 : POST_FAILURE in 42m 44s","accounts_in_message":[],"_revision_number":2},{"id":"4d7292f0f7544f587ef6e0ba84d6a0cf4097af5e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-04-14 15:44:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e3f0fa88cfbb48fc8c58bd4952b5cf84 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":2},{"id":"c10cc6976b7e839c9e609e0459de6a046ae834ac","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-04-14 15:52:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-cisco-zonemanager-job http://173.39.89.140/logs/12/880512/2/check/dsvm-tempest-cisco-zonemanager-job/fb65dc9/ : SUCCESS in 1h 04m 23s","accounts_in_message":[],"_revision_number":2},{"id":"80b3e442920f8b01d4965cb3d6390739f0eaa604","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-14 16:06:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/12/880512/2/check/iscsi-driver-alletra-6k/a86c91e/a86c91e.html : FAILURE in 1h 19m 58s","accounts_in_message":[],"_revision_number":2},{"id":"792ff99aa37b6f76aeded639e66df42fa49462a4","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-04-14 16:14:56.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/880512 : SUCCESS in 1h 26m 19s","accounts_in_message":[],"_revision_number":2},{"id":"29ed8469817f3c0913921c577a80a7dcf20354d2","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-04-14 16:21:45.000000000","message":"Patch Set 2:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/12/880512/2/check/fujitsu-eternus-dx-iscsi/d429005 : FAILURE in 1h 32m 14s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/12/880512/2/check/fujitsu-eternus-dx-fc/851734d : FAILURE in 1h 35m 08s","accounts_in_message":[],"_revision_number":2},{"id":"d0ab6f1b2722b4abac2be1012029103b59606aa5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-04-14 16:38:45.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/12/880512/2/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/114f3ab/ : FAILURE in 1h 44m 27s\n- pure-devstack-cinder-tempest-fc-aio-jammy pure-devstack-cinder-tempest-fc-aio-jammy : NODE_FAILURE in 0s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/12/880512/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/5e26988/ : SUCCESS in 1h 48m 06s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/12/880512/2/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/b360583/ : FAILURE in 29m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c5973a97379b6b5a94256c59bf4bc7a53cebb43a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 17:08:56.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d07b70caad649258a68dd21f6063057\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb1f2aa220e847df8de947ab13719ba8 : FAILURE in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/412b2792e7fe4c0e863a41f263865a30 : SUCCESS in 12m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6b02ee8ad574acab8fab612645570b8 : SUCCESS in 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26f9ce8ed8c44e7aa8d75eb8f2e73729 : SUCCESS in 11m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/1cd97393f62549bcbcfb542a0c6cdbe2 : SUCCESS in 1h 01m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8379c793412a49c98f5fc6089e3bd59b : SUCCESS in 1h 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a285013efef84ed6851dae9307762d4c : SUCCESS in 58m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/13039b85c615481b9d214558fc69a510 : SUCCESS in 16m 53s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/4a13cc74caa140a8a72fb8b3b9afe89b : SUCCESS in 4m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/abd0108f7efe42fcac6f871a5a30ee67 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/917e0e6bac944929b5dbd21d1c1a8d38 : SUCCESS in 5m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8c89919dfc1242dfad7907c0224ecdea : SUCCESS in 5m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/504a2cd42fe640139ac05ee48fd8739f : SUCCESS in 46m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3b2779473ddb4ae7a242437c9f7514f7 : SUCCESS in 6m 02s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6b228b209a0a4f498cd6054c25973ea1 : SUCCESS in 1h 32m 44s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/f7bf44b9c36f4e9e8c5949cb11bd2bf2 : FAILURE in 1h 24m 00s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/02176a7d20e4455181240c49fb973670 : FAILURE in 1h 44m 17s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f9c4b317ebed45c78177461bf2640db1 : SUCCESS in 1h 19m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/7218f1d20d84405dac4a9186ea5afe42 : SUCCESS in 1h 32m 10s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a614a06177c843b8841fb03522220fbf : FAILURE in 43m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/fc10f5d4270849f4a792c8de2539edff : SUCCESS in 1h 12m 38s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e81c79aca2014a009f551a0757246d80 : SUCCESS in 2h 22m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76091dfc4987465f84541c3cc6fe06c2 : SUCCESS in 56m 34s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3b16eba1a284171b1fddb8f4fa5a1d1 : SUCCESS in 1h 16m 04s","accounts_in_message":[],"_revision_number":2},{"id":"107c5e7cd7bf8242ab2d7fa2871904935aaf5f4e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-04-14 17:11:08.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-12-880512-1/FC/ : SUCCESS in 2h 18m 43s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-12-880512-1/iSCSI/ : SUCCESS in 2h 31m 25s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"18f38cc86f9d48fb938b061536225b888589691a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-04-14 17:19:36.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-12-880512-2/FC/ : SUCCESS in 2h 24m 58s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-12-880512-2/iSCSI/ : SUCCESS in 1h 51m 40s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"09b80167f275129a426c6a10d8b5261d9a3839e8","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-04-14 18:02:47.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/12/880512/2/ : FAILURE in 1h 22m 04s","accounts_in_message":[],"_revision_number":2},{"id":"65cefd77bb3ca754d8a074462220ac89f2e3eb18","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-04-14 18:08:10.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-12-880512-1/ns5_nfs : SUCCESS in 3h 25m 22s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-12-880512-1/ns5_iscsi : SUCCESS in 3h 28m 22s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"83fbe5207c5ed88690c5175cca4267416a8a46a2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-04-14 18:36:41.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-12-880512-2/ns5_nfs : SUCCESS in 3h 34m 18s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-12-880512-2/ns5_iscsi : SUCCESS in 3h 23m 56s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"6e610055de1b7e68eef10237b15ddbe66da387f9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-04-14 19:18:05.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/12/880512/2/check/DellEMC_Unity_FC/3894f87/DellEMC_Unity_FC/19 : SUCCESS in 1h 26m 06s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/12/880512/2/check/DellEMC_Unity_iSCSI/61080b5/DellEMC_Unity_iSCSI/19 : SUCCESS in 2h 06m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"34a221da657f97f609a1c1123948f1ff97f223a7","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-04-15 15:27:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/12/880512/2/20230415211334/NEC-V-ISCSI : SUCCESS in 2h 38m 19s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/12/880512/2/20230415211334/NEC-V-FC : SUCCESS in 3h 14m 14s","accounts_in_message":[],"_revision_number":2},{"id":"5e65cfc36e1073674df1f3b6ddec18d768f0a962","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-04-17 05:22:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/12/880512/2/20230417072359/NEC-ISCSI : SUCCESS in 3h 16m 15s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/12/880512/2/20230417072359/NEC-FC : SUCCESS in 6h 58m 33s","accounts_in_message":[],"_revision_number":2},{"id":"0bb0434cfcb2a535c8a9b3fd2dc25bda313eb5f3","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-20 12:48:27.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/12/880512/2/check/iscsi-driver-nimble/e35ac6c/e35ac6c.html : FAILURE in 1h 55m 20s","accounts_in_message":[],"_revision_number":2},{"id":"726ed35f618769291a0b0f2e98f60bcff7d13c00","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-08-26 17:20:13.000000000","message":"Abandoned\n\nHasn\u0027t been used since 2023.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2d251cb14cf430fa164bdc9a807f41c0a7edfe1a","revisions":{"cc59aed29fc0121072dface31a16718cca0aed47":{"kind":"REWORK","_number":1,"created":"2023-04-14 14:39:16.000000000","uploader":{"_account_id":35002,"deleted":true},"ref":"refs/changes/12/880512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/12/880512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/880512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/880512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/880512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/12/880512/1"}}},"commit":{"parents":[{"commit":"78fac2960d28e97f42454bb5f385b46fbae23a3c","subject":"Merge \"Shut up pylint about win32_disk_size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/78fac2960d28e97f42454bb5f385b46fbae23a3c"}]}],"author":{"name":"Luiz Santos","email":"luizfelipe.looze@gmail.com","date":"2023-04-14 14:37:11.000000000","tz":0},"committer":{"name":"Luiz Santos","email":"luizfelipe.looze@gmail.com","date":"2023-04-14 14:37:11.000000000","tz":0},"subject":"DNM - Netapp test to add upstream-check REST jobs with file filter to trigger only in Netapp driver","message":"DNM - Netapp test to add upstream-check REST jobs with file filter to trigger only in Netapp driver\n\nChange-Id: I41a1d83664fca194567d61c1b4eecf6a7a6a1334\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cc59aed29fc0121072dface31a16718cca0aed47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/cc59aed29fc0121072dface31a16718cca0aed47"}]},"branch":"refs/heads/master"},"2d251cb14cf430fa164bdc9a807f41c0a7edfe1a":{"kind":"REWORK","_number":2,"created":"2023-04-14 14:43:51.000000000","uploader":{"_account_id":35002,"deleted":true},"ref":"refs/changes/12/880512/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/12/880512/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/880512/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/880512/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/12/880512/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/12/880512/2"}}},"commit":{"parents":[{"commit":"78fac2960d28e97f42454bb5f385b46fbae23a3c","subject":"Merge \"Shut up pylint about win32_disk_size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/78fac2960d28e97f42454bb5f385b46fbae23a3c"}]}],"author":{"name":"Luiz Santos","email":"luizfelipe.looze@gmail.com","date":"2023-04-14 14:37:11.000000000","tz":0},"committer":{"name":"Luiz Santos","email":"luizfelipe.looze@gmail.com","date":"2023-04-14 14:43:28.000000000","tz":0},"subject":"DNM - Netapp test to add upstream-check REST jobs with file filter to trigger only in Netapp driver","message":"DNM - Netapp test to add upstream-check REST jobs with file filter to trigger only in Netapp driver\n\nChange-Id: I41a1d83664fca194567d61c1b4eecf6a7a6a1334\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2d251cb14cf430fa164bdc9a807f41c0a7edfe1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2d251cb14cf430fa164bdc9a807f41c0a7edfe1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
