)]}'
{"id":"openstack%2Fcinder~881032","triplet_id":"openstack%2Fcinder~master~Ibf2b50a23bff926a21c07ba2cc00ed16d956cf54","project":"openstack/cinder","branch":"master","topic":"rbd-backup-restore","attention_set":{"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2023-05-15 12:57:41.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2023-06-20 17:51:31.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2023-06-21 11:35:33.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ibf2b50a23bff926a21c07ba2cc00ed16d956cf54","subject":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak","status":"NEW","created":"2023-04-20 17:12:24.000000000","updated":"2023-07-04 02:06:22.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"121cca6b256f4aa2ec1deb2e8881923e37013e69","_number":881032,"virtual_id_number":881032,"owner":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-28 13:56:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-20 17:24:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-20 17:39:28.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-20 18:03:58.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-20 18:10:41.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-04-20 18:17:05.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-04-20 18:19:45.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-20 19:02:21.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-04-20 19:40:34.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-04-20 20:13:00.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-04-20 20:27:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-20 20:42:25.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-21 05:23:06.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-04-21 12:41:19.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-04-21 13:27:43.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-21 13:27:43.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-04-21 13:27:44.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-21 13:27:44.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-22 09:02:09.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2023-04-22 09:02:16.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-04-23 01:18:25.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-04-23 20:25:44.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-26 16:51:38.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-05-02 16:11:29.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2023-05-09 19:57:54.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2023-05-15 14:40:21.000000000","updated_by":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-20 14:26:58.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2023-06-21 04:12:04.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-06-27 14:05:57.000000000","updated_by":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"reviewer":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"state":"CC"},{"updated":"2023-06-27 18:26:21.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"}],"messages":[{"id":"4bf934fd3ea2efd109f1ad07756e75e505966527","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-04-20 17:12:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6bda10c7bbdc1cbeb63aebbc4e823978a98abb47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 17:24:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2fb921f2bf141958da0fde9b6b51f29\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/843998c6fcab4f9e889e547e3172f611 : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1a8e7b1811784e7fb43ade0efc1de3a9 : SUCCESS in 10m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a6c5ba97c45a74c93f41f10bc82aae37afdf70e7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-04-20 17:39:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/881032 : FAILURE in 25m 10s","accounts_in_message":[],"_revision_number":1},{"id":"216abe0173b8e383d5711d965d522f16668db39a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-04-20 18:03:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/32/881032/1/ : FAILURE in 43m 29s","accounts_in_message":[],"_revision_number":1},{"id":"74ce05872ac1d7169fdf6f0a7e2bc9a32a215312","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-04-20 18:10:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/58fe1dd1caf94df58617ea26d1936570 : FAILURE in 57m 19s","accounts_in_message":[],"_revision_number":1},{"id":"616bbc0724f62adac30963212a36e4eccc0219f6","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-04-20 18:17:05.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/b60bcf244a8e4aaa9a9032681453579b\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/ffbd556cdece48bfa814ecb2304fffd9 : POST_FAILURE in 1h 01m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8b5ff4ad1fe7c23967ccdf531ee7698459c7f01d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-04-20 18:19:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/d49f6dc8cbaa48f69fdf7ca5735cf350 : SUCCESS in 1h 05m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5f26dbb79a3ad9c8738f40d1ce6d739e24f2cc39","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-20 19:02:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/32/881032/1/check/iscsi-driver-alletra-6k/d6f33e6/d6f33e6.html : FAILURE in 1h 48m 32s","accounts_in_message":[],"_revision_number":1},{"id":"ee786cbb47a5d421b2c4f783789659ab44b3895a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-04-20 19:40:34.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-881032-1/FC/ : SUCCESS in 2h 27m 52s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-881032-1/iSCSI/ : SUCCESS in 1h 48m 10s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"29c0387715623dbb217431189b353912e372a9d7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-04-20 20:13:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy http://openstack-logs.purestorage.com/32/881032/1/thirdparty-check/pure-devstack-cinder-tempest-iscsi-aio-jammy/bd61e8b/ : SUCCESS in 1h 50m 43s\n- pure-devstack-cinder-tempest-fc-aio-jammy http://openstack-logs.purestorage.com/32/881032/1/thirdparty-check/pure-devstack-cinder-tempest-fc-aio-jammy/d5e2db7/ : FAILURE in 2h 26m 21s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy http://openstack-logs.purestorage.com/32/881032/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-roce-aio-jammy/05c8dac/ : SUCCESS in 1h 42m 55s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy http://openstack-logs.purestorage.com/32/881032/1/thirdparty-check/pure-devstack-cinder-tempest-nvme-tcp-aio-jammy/25213a0/ : FAILURE in 30m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb625a08209cfd6ad2fadd949df4b88a748fe7b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 20:27:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/127e34259a814f4c80293e9ebb7cd848\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76e5696a569e4cd8b8546c62633a05a1 : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa285c4a58d84d15855c912f679df485 : SUCCESS in 12m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/815168fabc044e658c528bf5f1cddeb0 : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99400d7ec8bc425a9300da399da83bc1 : SUCCESS in 12m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/1a19c4e7a97d4a83845660a062be1247 : FAILURE in 1h 04m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0af1eb640652446c97cc6b0f4352a6d8 : SUCCESS in 1h 14m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9176b5a0af0c40aca0cbc8471d5bddc4 : SUCCESS in 58m 55s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9ca174bf86a84081a0bb7d46d6b3f6cc : SUCCESS in 16m 43s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/92938d88931442e799fe276b3ae3670f : SUCCESS in 5m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/77837ac288d3466b956f0c6c1c79a148 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f4fc1256927457e87920829b932c11e : SUCCESS in 7m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f51ce7aba8ef4dfeb1aacb945f169fe4 : SUCCESS in 6m 37s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/e747652e2c0e4a72bf6a3c326d3f2686 : SUCCESS in 56m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/904488249d364390aaf5d3380554d3fb : SUCCESS in 6m 05s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/47849d6ce2c5433eb20c69d32a8a0bce : SUCCESS in 1h 09m 07s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/8f6441d7e0db4cacbb0de5a1bc34df15 : TIMED_OUT in 3h 06m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/84ad3661511545e3846579fd628af0e4 : SUCCESS in 1h 46m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/7edc271c630444c8bf959a7e3b11635c : SUCCESS in 1h 34m 56s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/62a44138aaaf48b38fc0e2bf0f32bded : SUCCESS in 1h 10m 19s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/a8f022d8569c47fba1dc9577e40daa36 : SUCCESS in 1h 14m 32s (non-voting)\n- cinder-tempest-ceph-multibackend-cbak https://zuul.opendev.org/t/openstack/build/8cf56fd1c434437dacafd6bd4c32933f : FAILURE in 1h 35m 05s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d569f8aaba554345b52da4857aa59cba : SUCCESS in 34m 27s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/9d2c42ba941b407fb7dcfa0d95569bcb : SUCCESS in 1h 22m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/bf61218813ca46849cdc22425f3c80b3 : SUCCESS in 1h 21m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/28fa52e496854437b542b03773fc4777 : SUCCESS in 2h 41m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/542758514a7845719d4e6451698fd835 : SUCCESS in 43m 13s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e1c592d429346c3903254cbc8bf33b3 : SUCCESS in 1h 06m 45s","accounts_in_message":[],"_revision_number":1},{"id":"55f0dc6c91f34be1e7d17e34de2d2eb5aeb694b8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-04-20 20:42:25.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-881032-1/ns5_nfs : SUCCESS in 3h 25m 28s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-881032-1/ns5_iscsi : SUCCESS in 3h 29m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"85a2aa89b917e69c5330a2dfeae2184531d73600","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-04-21 05:23:06.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/881032/1/check/DellEMC_Unity_FC/1bd6936/DellEMC_Unity_FC/53 : SUCCESS in 1h 35m 27s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/881032/1/check/DellEMC_Unity_iSCSI/ee59d0d/DellEMC_Unity_iSCSI/55 : SUCCESS in 2h 07m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"e0fd009c2fcd438b08fa58cc85d10b9490cf68eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-04-21 10:16:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"690e603458988f78c0ff6d0c449047e485724cd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 10:29:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/833a22259567445097f375e96a22f428\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/54a6c0d5c2034055acf765b84d7eb0a9 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2fd910e057e74309a3c4a68058d81a2a : SUCCESS in 11m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"69a5fb920f7ff6a1d2eff354662b9deb3b719a57","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-04-21 10:44:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/881032 : FAILURE in 25m 16s","accounts_in_message":[],"_revision_number":2},{"id":"b903da642ff925c9852aa0ee036c445abfd955bb","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-04-21 10:56:20.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/b849103cd5f74351891d60e7bdba05fd : POST_FAILURE in 38m 54s","accounts_in_message":[],"_revision_number":2},{"id":"9de83318c097fe4bff1c59ca874786b65dc5d9e0","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-04-21 11:12:14.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/32/881032/2/ : FAILURE in 47m 24s","accounts_in_message":[],"_revision_number":2},{"id":"ea03634c8e1e3619e1c9c6111959f0347098239f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-04-21 11:23:06.000000000","message":"Patch Set 2:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/16b73c40fe0042f6a0bf4b10a141b679\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/be564e8b39d54964ba971634e9dd8e5c : POST_FAILURE in 1h 03m 18s","accounts_in_message":[],"_revision_number":2},{"id":"e3bb5b927aa53d4b38de8c93570046e85bce7c39","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-21 11:50:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/32/881032/2/check/iscsi-driver-alletra-6k/b2605b4/b2605b4.html : FAILURE in 1h 32m 05s","accounts_in_message":[],"_revision_number":2},{"id":"85c9c05bf284993d5810ca24ff391181e16882d2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-04-21 12:30:54.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/881032/2/check/DellEMC_Unity_FC/633f15a/DellEMC_Unity_FC/55 : SUCCESS in 1h 33m 24s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/881032/2/check/DellEMC_Unity_iSCSI/e071879/DellEMC_Unity_iSCSI/57 : SUCCESS in 2h 14m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"4bd2f5a32acab7eff51223c8515a5451b508ccba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 12:39:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efdaaefcd2d94b72b00601fe4b5dd1f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c53b6eb589164b1f92b7175137954095 : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db9ea25c723a4d54b7a6b2ccd2b8e4e2 : SUCCESS in 15m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3685553252f4ebcba7ce52faad5ddd5 : SUCCESS in 14m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcb8e54a70884643bc6434e9846de6fb : SUCCESS in 13m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/88fbcec075794fe1be761daf88c22f71 : SUCCESS in 54m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/becacc789e05405a8398bbe51a979c98 : SUCCESS in 1h 21m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9766e5335b0d41c9b0ce899688456719 : SUCCESS in 59m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9642f47e3e9a49a6a8caba692032ea8c : SUCCESS in 17m 24s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/83afff08558745c5ab332894cccb2e79 : SUCCESS in 6m 44s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e1b79e8bf6984e93a5b2a2ec23e74dfc : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7c34790bdd3e44f8b3d48e6b5f6652b8 : SUCCESS in 6m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bcfdb693b93e4034b451fb7da74ab0c0 : SUCCESS in 5m 24s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/2e3f031bdfa84e189b289cd79483071e : SUCCESS in 42m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6c112bb05a8f45139c00d5e3afa02393 : SUCCESS in 6m 56s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b7eb024817764171be78a91ca2e89915 : SUCCESS in 1h 14m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0642a29635b44313b913bc070c4d963b : FAILURE in 1h 34m 07s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/e1ed23155fc043388b203058595dcdf9 : SUCCESS in 1h 53m 59s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/470255c35ad346fc97d1f151e074cc20 : SUCCESS in 1h 56m 36s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/26f8e055c45c40cf95b82032a581c104 : SUCCESS in 1h 12m 05s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/6d2a3971718845a2875ef4213c1c020a : SUCCESS in 1h 33m 35s (non-voting)\n- cinder-tempest-ceph-multibackend-cbak https://zuul.opendev.org/t/openstack/build/8dc8aa24d6e744709c8af2d6dfd38c51 : SUCCESS in 1h 32m 54s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/931456556047451b8f651dcb89a1e07f : SUCCESS in 41m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/93174ecb68aa43a8a4e0933726b7b23b : SUCCESS in 1h 19m 21s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1e477bfa573c41c9b3f9a4daf16c449d : SUCCESS in 1h 10m 09s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/79bd14a923a9438ab7d4b98ef7d633ee : SUCCESS in 2h 17m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/152dc42945a24f6096b52d4a9a31f325 : SUCCESS in 50m 45s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5048eefccaa6403cbcafbc5eb22217a5 : SUCCESS in 1h 35m 40s","accounts_in_message":[],"_revision_number":2},{"id":"40627c4017df7d0e27655550ede90bb9f75fcd37","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-04-21 12:41:19.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"a3042560bc8efffde560ce1afecbb4141e68d351","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-04-21 12:47:33.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-881032-2/FC/ : SUCCESS in 2h 30m 43s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-881032-2/iSCSI/ : SUCCESS in 2h 22m 15s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"c0387ceb069ef6b25c67f336a3554208d4f27bd4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-04-21 13:27:43.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"0934724f7fea7e0d072ff1afcff191794559bff3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-04-21 13:27:43.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"5b7b0152848217146410a635d67cb4a752727d0a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-04-21 13:27:44.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"297b69f402638cacc5243638f85230b667931022","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-04-21 13:27:44.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"8c30ec149616df6692bea50147435aa528c2cc14","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-04-21 13:27:44.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"74584f4b87cbb725ade2fa646e184008371768d7","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-04-21 13:27:58.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"f9f3d80213da56c4e79d7e64015d02f2457d9dc9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-21 13:27:58.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"f5e3810c20e5f58a57440e9fd4da0e2b217bfb64","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-04-21 13:45:36.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-881032-2/ns5_nfs : SUCCESS in 3h 24m 55s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-881032-2/ns5_iscsi : SUCCESS in 3h 28m 37s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"8785d90c112369ca552ded6e9a3afbc61a734828","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-04-22 09:02:09.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"162001fda1fc09ab06f0d72a2e2c08685f66ca92","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-04-22 09:02:16.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"948fd93794b5c3b01a1474edc32f55c8e5496166","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-04-23 01:18:25.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/881032/2/20230423072313/NEC-V-ISCSI : SUCCESS in 2h 23m 17s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/881032/2/20230423072313/NEC-V-FC : SUCCESS in 2h 55m 09s","accounts_in_message":[],"_revision_number":2},{"id":"2d2e167f3e743565baa7661da904e931dbd6ae1f","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-04-23 20:25:44.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/881032/2/20230424020703/NEC-FC : SUCCESS in 3h 16m 42s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/881032/2/20230424020703/NEC-ISCSI : SUCCESS in 3h 18m 38s","accounts_in_message":[],"_revision_number":2},{"id":"e1c43024549c8ab2a516bd86ad56b0916e42006e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-04-26 11:03:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d23e215e6d48de68ded726364e929967e956c953","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 11:16:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d51c1366a36e4bcca6239687e158ff5c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c956ce5a09d4719a37bab6bb0e59275 : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c5af3b1c1dca404cb001eb30f50afa2b : SUCCESS in 11m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"90c0a7d08b96e0e936194617e37cc9f24125acbe","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-04-26 12:18:04.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/142459f818d544328cc5534d2eb023d9 : POST_FAILURE in 36m 03s","accounts_in_message":[],"_revision_number":3},{"id":"b0a1494fa7f0df599d3bd6fd616e953f88e55d9a","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-04-26 12:34:38.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/32/881032/3/ : POST_FAILURE in 44m 04s","accounts_in_message":[],"_revision_number":3},{"id":"257c92829777bbdab263394c8fd16ec9adb593cc","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-26 12:52:23.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/32/881032/3/check/iscsi-driver-alletra-6k/c0e51ca/c0e51ca.html : FAILURE in 1h 47m 02s","accounts_in_message":[],"_revision_number":3},{"id":"338f58d7a56a046bd925ccc69321aad42a8965c8","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-04-26 12:52:56.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/881032 : SUCCESS in 1h 23m 29s","accounts_in_message":[],"_revision_number":3},{"id":"bdbaaf4c0c9a306bbdb58f11a3f253087864c4e2","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-04-26 13:25:09.000000000","message":"Patch Set 3:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/193055647137455990d19d0a0cf8c657\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/e639d143cb244fde9cf04f9d5dc47e4d : SUCCESS in 1h 25m 25s","accounts_in_message":[],"_revision_number":3},{"id":"dbf5ac045808e1b2d9ac2dd438cdfd6746051d81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 13:42:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ace7596e2bbf4953882daea8002b14a7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/978f5e31e1a54e4bae81223de12e8ff0 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/385108053e784d35aaed7245b5756af2 : SUCCESS in 11m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3759d1b5f2c74d4db92c1274d06a59f9 : SUCCESS in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57f384faf50442bfbed01b0ff906b813 : SUCCESS in 12m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/e1b3ae7cac0f442d98b5024541dfe4f5 : SUCCESS in 46m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0ac3ee738b3744bc94d4ab30aa3feda1 : SUCCESS in 1h 14m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f75cf821d7824e03beeef438cbd7fcdb : SUCCESS in 59m 38s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/92e9fa898e50422d86d1da4c2e692146 : SUCCESS in 17m 14s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/cb11462ac265454e9e4b7cb9eebd4b3a : SUCCESS in 5m 25s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/461cbd19c6494874956e09a93ffb5d51 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/33ff37e835364687a94be05a2ef67a86 : SUCCESS in 6m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cc02c10871794e04ab92d34f52be74e0 : SUCCESS in 5m 52s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/37d47d276ae345ac8426b85783e41759 : SUCCESS in 51m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d9df531facdc40c8816b674ba5b64413 : SUCCESS in 7m 30s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fbd268eb82404f32837d028854f7272c : SUCCESS in 1h 18m 29s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d888336dafe54bf496320be21337e62b : FAILURE in 1h 24m 58s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a449dc9d4b0b4daa83f6c356024d60b0 : SUCCESS in 1h 30m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/18bb062b01144d9c8f37b79b7184c78f : SUCCESS in 1h 52m 37s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/e48d680f8f1a4588b5e336831cc4c346 : SUCCESS in 1h 02m 50s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/94b7d898e9234c5cb6fc6d434fc26b1e : SUCCESS in 1h 46m 23s (non-voting)\n- cinder-tempest-ceph-multibackend-cbak https://zuul.opendev.org/t/openstack/build/d0f6e9a1c0bc4c91b2e1173098e11256 : SUCCESS in 1h 45m 17s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/46befd2736f8409eb82ff9082cae4d88 : SUCCESS in 43m 30s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d5dcea224f6644c38b1d8444b21cc6bb : SUCCESS in 1h 20m 43s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/cc50a18fbe474ccdb1c09c8d970ff6ed : SUCCESS in 1h 18m 30s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2097d17922834fb49992e05f80a48567 : SUCCESS in 2h 30m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f7267af691a458197558527df4b0af0 : SUCCESS in 58m 32s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba25c5579b5e41928b199b65f12a4d9f : SUCCESS in 1h 18m 39s","accounts_in_message":[],"_revision_number":3},{"id":"4458f49427d861c28468c96415efc1d3d97901ad","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-04-26 13:44:21.000000000","message":"Patch Set 3:\n\nBuild failed.","accounts_in_message":[],"_revision_number":3},{"id":"6bd9bd5ca78aff31ec1f4253c6ae8c4b36b7080f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-04-26 15:07:16.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-881032-3/ns5_nfs : SUCCESS in 3h 34m 44s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-881032-3/ns5_iscsi : SUCCESS in 3h 39m 16s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"49cae07218ec14f49d0ef33fb1f0b8f8a8fce364","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-04-26 15:07:51.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/881032/3/upstream-check/cinder-tempest-plugin-ontap-fcp/f6c42e4/ : SUCCESS in 1h 24m 14s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/881032/3/upstream-check/cinder-tempest-plugin-ontap-iscsi/870beee/ : SUCCESS in 1h 11m 26s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/881032/3/upstream-check/cinder-tempest-plugin-ontap-nfs/fd88b18/ : SUCCESS in 42m 31s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/881032/3/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/4c56837/ : FAILURE in 41m 58s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"19890e94571f43c8a45f31c72d45a7c0a17d5768","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-04-26 15:18:57.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/881032/3/check/DellEMC_Unity_FC/5b9d12b/DellEMC_Unity_FC/73 : SUCCESS in 1h 36m 04s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/881032/3/check/DellEMC_Unity_iSCSI/02992fb/DellEMC_Unity_iSCSI/76 : SUCCESS in 2h 02m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"6082e24091f6066025a23c94cba5bc6f0635bb98","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-04-26 16:19:38.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/32/881032/3/check/DellEMC_PowerFlex/46c800f/DellEMC_PowerFlex/44 : SUCCESS in 1h 20m 48s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"e1f2fbb29354f9ca2943d62f488a150144c20cc3","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-04-26 16:51:38.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/32/881032/3/check/DellEMC_PowerStore_iSCSI/a25c8f9/DellEMC_PowerStore_iSCSI/5 : SUCCESS in 1h 13m 14s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/32/881032/3/check/DellEMC_PowerStore_FC/f91648d/DellEMC_PowerStore_FC/6 : SUCCESS in 1h 29m 38s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/32/881032/3/check/DellEMC_PowerStore_NFS/5a1033a/DellEMC_PowerStore_NFS/4 : SUCCESS in 1h 15m 55s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/32/881032/3/check/DellEMC_PowerStore_NVMeTCP/ef9fab4/DellEMC_PowerStore_NVMeTCP/4 : SUCCESS in 1h 29m 53s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"52c60145adbc3e0d73b7ecf86d8030b8123f9cda","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-04-26 19:41:15.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/881032/3/20230426233449/NEC-V-FC : SUCCESS in 2h 57m 30s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/881032/3/20230426233449/NEC-V-ISCSI : SUCCESS in 5h 06m 22s","accounts_in_message":[],"_revision_number":3},{"id":"54f2c56a39c3f6cf175872626cfd4e1c3d23ad98","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-04-27 01:41:30.000000000","message":"Patch Set 3:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/881032/3/20230427065305/NEC-ISCSI : SUCCESS in 3h 26m 26s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/881032/3/20230427065305/NEC-FC : FAILURE in 3h 48m 22s","accounts_in_message":[],"_revision_number":3},{"id":"7bbccf9c8e922033bdc6899648016ab0d4722211","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-04-27 01:49:50.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/881032-3/PVME-iSCSI/ : SUCCESS in 1:49:21\n- PVME-FC https://os-logs.seagate.com/dell-ci/881032-3/PVME-FC/ : SUCCESS in 1:25:46","accounts_in_message":[],"_revision_number":3},{"id":"08dcfe1b4e1caff3307b4e514653d6c9024b0f95","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-04-27 01:53:46.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/881032-3/Seagate-iSCSI/ : SUCCESS in 1:47:30\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/881032-3/Seagate-FC/ : SUCCESS in 1:25:16","accounts_in_message":[],"_revision_number":3},{"id":"d6fa60cc6ba72d86ed433367a1230607c0b1f82a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-04-27 06:43:41.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/32/881032/3/check/DellEMC_PowerMAX_FC/e70fb47/DellEMC_PowerMAX_FC/32 : SUCCESS in 6h 30m 56s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/32/881032/3/check/DellEMC_PowerMAX_iSCSI/0850222/DellEMC_PowerMAX_iSCSI/31 : SUCCESS in 6h 11m 48s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"3243f7cc9f51c5e5b6127e77807ec9e71395ac2b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-04-28 11:11:02.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/340f089255ef4958a4599b2e682c80bc : RETRY_LIMIT in 1m 00s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/5495878bf7ff4b18ab296cda3eb1d034 : RETRY_LIMIT in 1m 02s","accounts_in_message":[],"_revision_number":3},{"id":"63150a403e1b3ef0fe4300701fe9265af8ba1e58","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2023-05-02 16:11:29.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4c9bdaa67558f9871854df264108e88029606972","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-05-09 19:57:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5d46578e584cdccc1638ac04ca2e9778a567a205","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-10 09:28:21.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"3e4c6a1cd1bb47f8832c709a30fc97b150005f91","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-05-10 10:31:47.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"73f7aa8a27ad1b9ada9dcd29b1706ca03a2caacf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-05-15 12:57:41.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"42eb57a15df8be111da20addab2637d2d852d9a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 13:11:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce3e278cff9947949ac66f82926a1160\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0a4adc611464b2882248897d9a65aa0 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba2180576b3c4cd095b8134e58bae7aa : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c0c22ceac7a6480a8dfcd0f2cee21058 : SUCCESS in 12m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a8a9d4f496ac5c4e933ba5fd538ca67f7cfe9575","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-05-15 14:01:27.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/881032 : FAILURE in 1h 01m 43s","accounts_in_message":[],"_revision_number":4},{"id":"5246876040016a4758cc1b080442590969a41f1e","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-15 14:30:30.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/085fe07cb1604a29a43afcc9ae292e48\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/be636a3a5c054ebd8621775d37bc81ff : FAILURE in 1h 29m 43s","accounts_in_message":[],"_revision_number":4},{"id":"6b6a5231c97c119770f5330f6b2a03525f9ee874","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-05-15 14:40:21.000000000","message":"Patch Set 4:\n\nBuild failed.\n - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n - To recheck, add \u0027run-Fujitsu ETERNUS CI\u0027 or \u0027recheck\u0027 to your comment.\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/32/881032/4/check/fujitsu-eternus-dx-iscsi/dac1235 : FAILURE in 1h 39m 45s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/32/881032/4/check/fujitsu-eternus-dx-fc/8a089af : FAILURE in 1h 41m 53s","accounts_in_message":[],"_revision_number":4},{"id":"a8ab1d4836cb9d3a618a2840725ec254643e4209","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-15 14:56:41.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/32/881032/4/check/iscsi-driver-alletra-6k/b69e3b2/b69e3b2.html : FAILURE in 1h 57m 15s","accounts_in_message":[],"_revision_number":4},{"id":"92ab4523c771895eb0a65e65a28d83ef0ee2df8e","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-15 14:59:58.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/f85fa78640f74a1c9b028bae6aa2be11 : TIMED_OUT in 2h 01m 12s","accounts_in_message":[],"_revision_number":4},{"id":"68ff75ae18d6a7ecc2dd90b4aa1ab33b2a40e1cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 15:34:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ce812158ca2445a94d78972695e374b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afc53c95787b48318ec639165a08c433 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f64ddb0bb8814a5f93009e87389b11dd : SUCCESS in 11m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0355b97996f40a49de9a45540027009 : SUCCESS in 11m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d5b77b1e2d242f5b9965bb4256815fc : FAILURE in 11m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f91616a018440d18a68c10516c414cd : SUCCESS in 13m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/a03ff35c38c64f01b3bbf995977b8389 : SUCCESS in 54m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c5214d85c2af48daa7eec60a734271b4 : SUCCESS in 1h 34m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d812a334856d41c7896af3e2b8ff5ff1 : SUCCESS in 52m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/e7d0840a84a74f14ab1cdbd7217db03f : SUCCESS in 16m 52s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c585f05a60f6492288c3c67451222c3a : SUCCESS in 5m 01s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a0ad64a9142d4b89bce5be91d337241b : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2ca6c609d2044cf18884440f0c97159e : SUCCESS in 5m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/59e1577e6179419dab4f775be62364be : SUCCESS in 5m 39s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0d8c7b60faf24f7ab86887cdd50d1f7a : SUCCESS in 36m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7ec29abdb9544beeb3198674223b035a : SUCCESS in 5m 42s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9703d0fac77440f5b638df85354bcdab : SUCCESS in 1h 54m 26s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/bcf72fded39648d3a857bfec7d7a7372 : FAILURE in 1h 49m 57s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/167ed37abb78402aabf12868eb8af529 : SUCCESS in 1h 47m 25s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/ee58afa9267141809a00ffbe2025f65c : SUCCESS in 1h 47m 22s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2cffb5f2545840ba8e82547256bdca53 : SUCCESS in 1h 08m 47s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/3d75e8f4271c4031960e8695ef54a777 : SUCCESS in 1h 31m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/fc39ee7a14b748ef98413f85c5f779e8 : SUCCESS in 48m 00s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/275d1f6487b249429ca7cadfe868ee13 : SUCCESS in 1h 23m 58s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/034c58e415bf4eeb8a9f196e0488bb07 : SUCCESS in 1h 21m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f2d51cab88ce4877ac708984f9c5eea4 : SUCCESS in 2h 30m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad9db171a40241e180bcd85385ddd5ed : SUCCESS in 50m 34s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ab145d77ff94835a98dc1987de0e8c2 : SUCCESS in 1h 34m 02s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/274c2e37090841678a3d2b6f0ce97b19 : SUCCESS in 9m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fa3289b050cc8f3ee74aed563716a5959bdea9e2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-05-15 16:55:42.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-881032-4/ns5_nfs : SUCCESS in 3h 53m 57s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-881032-4/ns5_iscsi : SUCCESS in 3h 57m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"6a1a1dc9b74eeadf6f8707a2ec46b79388a0a41b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-15 16:59:40.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/881032/4/check/DellEMC_Unity_FC/e8e3bb4/DellEMC_Unity_FC/142 : SUCCESS in 1h 46m 37s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/881032/4/check/DellEMC_Unity_iSCSI/17bd19f/DellEMC_Unity_iSCSI/145 : SUCCESS in 1h 54m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"efe6bb1928926c498d2e4b045eadeebf4ac5dc76","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-05-15 23:56:48.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/881032/4/20230516061520/NEC-V-ISCSI : SUCCESS in 2h 20m 33s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/881032/4/20230516061520/NEC-V-FC : SUCCESS in 2h 41m 25s","accounts_in_message":[],"_revision_number":4},{"id":"40b13b6e361ae447ba735711e73720ffb3dd17f4","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-05-16 07:52:03.000000000","message":"Patch Set 4:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/881032/4/20230516055736/NEC-FC : SUCCESS in 6h 58m 32s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/881032/4/20230516055736/NEC-ISCSI : FAILURE in 10h 54m 24s","accounts_in_message":[],"_revision_number":4},{"id":"3135982046a8ba692deb3c9c5d9a363232b7ced5","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-05-16 11:24:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ae6b7c121ae3f9baa9f367b95fb978940fd138f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 11:53:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afa0f5526ca340d4abc15414a606cc82\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/703c33b4e43345f0b7dfa55e106007aa : SUCCESS in 25m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e6e8a9dc6ec40b1843f62c6bdc829f2 : SUCCESS in 25m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/218a75c9380c42f288eb068b73639f58 : SUCCESS in 8m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e10c69359f9b7eb94aa7e0507e24560f4fdeca87","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-16 12:07:57.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/f8783aa1dd084670a67864bd03ba2b49 : RETRY_LIMIT in 13m 33s","accounts_in_message":[],"_revision_number":4},{"id":"a5fa3d6ddc1f7c0e4aea8df9407a65a872d294ee","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-16 12:33:13.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/6edd91dda92c4f959312ff74f27c43d2 : POST_FAILURE in 51m 09s","accounts_in_message":[],"_revision_number":4},{"id":"0e5836cd09bffca307601971cf4bb3f73da4b9e4","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-16 12:35:33.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/504d8123fe1849cd9bd9082837da3238\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/62749ede6eeb4c25b61bfbbc3f616b4b : FAILURE in 1h 08m 20s","accounts_in_message":[],"_revision_number":4},{"id":"bdfb9f7e1760467e0898268aab9b42bc8fcdb970","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 14:13:57.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/29cd1a71bd0c4bdd9f0332c326ef487b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4fe5ee080f6437d914233d72f46eae1 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/286b7d9848cb4feca5a5a5d380795229 : SUCCESS in 10m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3af50e7bc92f40e299b7225f75754c30 : SUCCESS in 11m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6120592f56e463992b1687cf9e1e40a : SUCCESS in 16m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66415783aec04ad1bfd0f444f2932548 : SUCCESS in 14m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/56e895730af846d39b74ed46214c1151 : SUCCESS in 1h 04m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/412b49269bc1495fa3259a5cbbe0d104 : SUCCESS in 1h 28m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6253ab37308e4639b81d9a31a3636be5 : SUCCESS in 57m 10s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/256b8bb4d66540878fb13d5fd53084a3 : SUCCESS in 18m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/406eab9c85394c559499a373ce3fff09 : SUCCESS in 5m 13s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e3de363c02d24d30b345295b466c5019 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aeaba3d920f840de85378e0a3b7cc6ea : SUCCESS in 6m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/558eb945a50d41999aebd74e7e512aaa : SUCCESS in 5m 43s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/b0794197040a40b5a99fcb370ff71940 : SUCCESS in 34m 39s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7b256e1bb82c4c859286490fed8d0834 : SUCCESS in 5m 59s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/61413c4e080c4c5e9df071b4e3bcffe4 : FAILURE in 1h 45m 53s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d9ed9d4dbddb4a61a19ff06759022c92 : FAILURE in 2h 07m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2f6ad2311e4646df91c01a0885aba86f : SUCCESS in 1h 59m 53s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/fdbdff2da17b4db785128495c31df429 : SUCCESS in 2h 22m 05s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/682686fed4cd4679b543b15cd3bdcfc8 : SUCCESS in 1h 03m 25s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9fb263bb80b7459e8fa2fe78856b4250 : FAILURE in 1h 37m 23s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/7c87474049814eb58db53e67d56f39f5 : SUCCESS in 42m 54s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/445eda9e4bc3427c8bdfcf0a533e9881 : SUCCESS in 1h 14m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e51ffdb948b649bbb0ee80b36a356df3 : SUCCESS in 1h 36m 47s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e7d6bc0fc7844f4d986a615e01162ed1 : SUCCESS in 2h 40m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c151cbe4d94499397d8dfe478d0f43f : SUCCESS in 46m 14s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebf658b59021432bb746cd5ff5fd202b : SUCCESS in 1h 17m 21s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/2f85b33e212e46438dd8aa1af4662a14 : SUCCESS in 11m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cbc6db1b68d37c9e42ba3825afe0dcae5947158e","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-05-16 16:17:38.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/32/881032/4/ : RETRY_LIMIT in 11m 12s","accounts_in_message":[],"_revision_number":4},{"id":"c2fe7d52137c0c2f20c54a83ba0986078938e970","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-05-18 09:40:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d2b6f858bc154ba3c202fc54dde806a83cfff3e5","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-05-18 09:41:17.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"48656966fb1cb56c2d79e7f9065a22a24e7608b5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-18 09:42:47.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/ec99df3a25bd4ae586b31f3920a98e95 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"0f760d30b2085f34663195490d11c578d4f27cd7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 09:51:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5d37ec3570549b5b2a6459ce6151258\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/56cda45f9a0747c59e3123731d84e1f1 : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b89d5973fa5f4aceb5d2864fa0e974c6 : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3400d7ddbf3f4bd7aa0fdb8fd7658927 : SUCCESS in 8m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"158c449cc67a75565265cce3ef286172330fdfe5","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-18 10:28:54.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/fe7465a11c27409fa72f11deabc0c255 : POST_FAILURE in 47m 05s","accounts_in_message":[],"_revision_number":4},{"id":"3b7043d553f1876b42b6d8ea15cab0bc08025782","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-18 10:37:31.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/ed117450547e4c10adaac9f7ff709552\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/85252cdc82ff46f4962cd9e24606d124 : FAILURE in 53m 01s","accounts_in_message":[],"_revision_number":4},{"id":"85ff55cb1dcf77bb6d04ea7ab5ed72b0ad5deabc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 12:05:37.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b0502c6d879470c972110042db6d8c6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f40ad2cfa22744ed89784c76ba4eee92 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e70ae33b682e4bab876530215f107c52 : SUCCESS in 13m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aaa3b2907f6b4f98a29b7fbe298ed4b2 : SUCCESS in 11m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4bf5335ab783482b9fb4dae7ea7377cc : SUCCESS in 13m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4adfdcfa810b481d85d386237ea3ad99 : SUCCESS in 13m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/3bbeba7d26574dada9707ce7e3fc8786 : FAILURE in 55m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/48d5f0ca51e1429bbe79d9e9f955392d : SUCCESS in 1h 26m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f7cb617e71674379ad46f0b25ba7718c : SUCCESS in 57m 03s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0f54eb6c7c3a43f9b378cb2438e8763f : SUCCESS in 17m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9bb36f99d80e4a399b42edf76757542c : SUCCESS in 4m 59s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/efce8e19dc564df2941b58718b0eb6e8 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e495f131e4b8400da14d43d45f6a5f97 : SUCCESS in 5m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b514ec569acc4416a0b1cff03f6cb24f : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/9ab7a177e8a340a5ba6509fb8f12246b : SUCCESS in 39m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8889959ef3274dba867577cb8b9dd975 : SUCCESS in 6m 53s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b4ffd85f3f174571bbe56a90f5ff2279 : SUCCESS in 1h 32m 16s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/6e910e26ecf8439c950779e43004f8ed : FAILURE in 1h 30m 37s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/236cfa57598f4b2b98d1144d02defc49 : SUCCESS in 1h 46m 50s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/5e33a6f6bb34444b962e7fe552c1ea04 : POST_FAILURE in 11m 54s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/6a0d6bde26f443d983d4d6bb2dd0dd2c : SUCCESS in 55m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ff9bd8cb2bf044a69655cf2ff630d150 : SUCCESS in 1h 28m 11s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/128f2d3c7d334b958a39a7cd387f687d : SUCCESS in 49m 12s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/66d16605d08b4fac829dbfb2e4a23e45 : SUCCESS in 1h 12m 31s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/1d6dac17baaa46239fa1ab9444444df8 : FAILURE in 7m 41s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f43eab3a3f504db59c1322d6483c99fa : SUCCESS in 2h 19m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/326fb316a4ec44cabf4a3bdb28dddb92 : SUCCESS in 58m 57s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4b9bd26ecc7452dbc4650a14878eb89 : SUCCESS in 1h 25m 56s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/46c32963f3ec4271bf4aba58a3aa98b3 : SUCCESS in 12m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b3e193c63a507153227817f6a15d3d2abaaf10fa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-05-20 05:40:56.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/32/881032/4/check/tempest-dsvm-full-focal-py3/6f8c88d/ : FAILURE in 1h 18m 33s\n- devstack-plugin-ceph-dsvm-focal-py3 https://ftp2.osuosl.org/pub/ppc64el/openstack/cinder/32/881032/4/check/devstack-plugin-ceph-dsvm-focal-py3/7d0427e/ : FAILURE in 1h 20m 59s","accounts_in_message":[],"_revision_number":4},{"id":"34357e8be1350c0e3d54353777c6fe4f77fee3d5","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-05-22 09:39:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7a2ced9951aaeb37f3c434e06cbe06efdbf22019","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-05-22 09:39:48.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"c79d96ce6c0510826c3bef7774dca126c017a9ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 10:18:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4583326651a0455da3c2ca0aa127e305\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a3d4e3d5cfe4aa88229a97f4c053250 : FAILURE in 30m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6558e7da2232466b9d16789a112f7b20 : FAILURE in 36m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b88fc08432a54b578b8bd4817f60cb1e : FAILURE in 27m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3da164af812e7a604cfd99ab94e2ba621b7f11d9","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-22 10:28:09.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/0f69613c58e04d3e913d422d6a295c46 : POST_FAILURE in 47m 42s","accounts_in_message":[],"_revision_number":4},{"id":"7fc148fd73b56e88ef0eab8dede831f1243557ff","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-22 10:41:00.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/d16673446bf545ba8522d746434745a0\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/043c8bdba79e40a9a1b3d4ce7b327288 : FAILURE in 53m 03s","accounts_in_message":[],"_revision_number":4},{"id":"d6785376205bf0ba3d3817a1d252a5e4c44f0778","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-05-22 10:41:10.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/a1def20b99d94d47bfc12db78a791a2d : SUCCESS in 59m 56s","accounts_in_message":[],"_revision_number":4},{"id":"e75c460880ddea7d4fa720843b66143bca2d2b42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 11:54:17.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a482a7f5e674eed80a4559e07839068\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcedfb03c66e4607985d6f818ae395ac : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62e15aaa4fe14a8e92bd366b95e4b17c : SUCCESS in 13m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6de8f2cfaad1434c85fd115e612345a0 : SUCCESS in 12m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11eed7a8b09f4c4fa0e063f2c2dc3945 : SUCCESS in 10m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6dc6ae29d8f4e1b891bd4ffa3fa3abb : SUCCESS in 14m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/a23d7dbf88a84b65bc66e60ba5ec0979 : SUCCESS in 51m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/53475db963ef47f8b36f08312586e5aa : SUCCESS in 1h 38m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/feb12e69df4c4053965072aac497e010 : SUCCESS in 55m 18s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9d80752f4b754a5681cff1f871fc0d8e : SUCCESS in 16m 23s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/05fe57f5fc70438ea8a07d434092f23b : SUCCESS in 5m 24s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3744142311b34c82bc29b39c52308784 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/882384a1b8be4a7da79b0bba1e377cca : SUCCESS in 6m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2cef5ccbcd8f4d70897688a9e5607e11 : SUCCESS in 5m 45s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f0c1917b73fb45ef9798caac1056b598 : SUCCESS in 48m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6e086b903c8d4df99367f31eb7026ba6 : SUCCESS in 5m 57s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/68c7822d299747c98c87550563fbb363 : SUCCESS in 1h 48m 15s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/344bec91f6c845c9b0c4cf061c94fc52 : FAILURE in 1h 41m 53s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d959628485bf4c29a2e2932bad346c0d : SUCCESS in 2h 07m 26s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e3d534400b03474c89da87352f9b3e2e : POST_FAILURE in 19m 41s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/38c4806875ca4d1fb85aaa6ee55e1afd : FAILURE in 58m 54s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/715ba72cc9b44677b3c6fd7509b957f0 : SUCCESS in 1h 47m 28s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/d6f9b5fb148c4f5d8db00128fb9542d7 : SUCCESS in 48m 18s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1cfe686718684e2fa5fad1fa2074f4a6 : SUCCESS in 1h 24m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5a11325ce3f14724ad9714bfc2bc5083 : FAILURE in 21m 19s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e1574315c37f4912b6350809bd7741e8 : SUCCESS in 1h 59m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cb6c92963454e7aba5b253d352bc764 : SUCCESS in 55m 58s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/504a5ec1496449c28dcb03cf67fc0393 : SUCCESS in 1h 16m 53s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/30271618b3ab4750b3afdcd0c4cad899 : SUCCESS in 9m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"42ade68fb5110e762e3a6cc2dd1c03dbc5e5db46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-06-15 11:48:42.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e6cfde784226751e5aa1bd6b578757654f10831f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 11:49:02.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 881032,5","accounts_in_message":[],"_revision_number":5},{"id":"0748b5811c7e1a411033dddf8e8bb2cc50341f90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 11:49:04.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 881032,5","accounts_in_message":[],"_revision_number":5},{"id":"50fb579b500bc410e41c56087c429b449c570c5b","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-06-15 11:49:21.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"65b58550109af5746ffbde156509156963342b73","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-15 11:49:24.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"efce4e092324612c9677618ec5d81923b2e1a79d","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-15 11:49:30.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/cinder for 881032,5","accounts_in_message":[],"_revision_number":5},{"id":"b2481e910f649888bb155ff554b6a4d1aa86fac7","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-15 11:49:42.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"e71e3fad448eb3a0d06481a55b19eab803f91678","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-15 11:49:42.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"a331a3518bdd69d3cfe4cb78d7647279645888ed","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-06-15 11:50:03.000000000","message":"Patch Set 5: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"98003e020985873f36a4e0b66bfc1e3f05ddc8b7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-15 16:25:36.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-881032-5/ns5_nfs : FAILURE in 4h 9m 44s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-881032-5/ns5_iscsi : FAILURE in 4h 25m 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"1394949258519ab8becffae271cc671e3bcb4e10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-06-15 17:56:40.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":6},{"id":"4d61fe3c0fcfea0fcfc3c98541d8143ac1492d5b","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-15 17:57:33.000000000","message":"Patch Set 6:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1c987fb019b1b31bb001ede37a6d975d3a0b142c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 18:09:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23285e6c3ab244be956bbfa4d24f50b8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cde2d818e08c4f0daef8a2b26d6c3f36 : SUCCESS in 12m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b8665329c014da88985e00b9a1841e9 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/26c7214381c14f43917f7f575674e1f6 : SUCCESS in 8m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f80f52d9a7293a9c02bd0f9a273300960da3c908","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-15 18:36:50.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/881032 : FAILURE in 37m 47s","accounts_in_message":[],"_revision_number":6},{"id":"bd791e6b29ba15e780104b921d71a26ce74a230f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-15 19:19:43.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/32/881032/6/check/iscsi-driver-alletra-6k/6c24d8c/6c24d8c.html : FAILURE in 1h 19m 54s","accounts_in_message":[],"_revision_number":6},{"id":"6cae64c7f0245be2862f73992e184e018584796b","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-15 20:21:51.000000000","message":"Patch Set 6:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/d1c1126b31b146b69e78169c0afe3696\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/53e5f296f97d49afb07bb19f5d03696f : SUCCESS in 1h 14m 55s","accounts_in_message":[],"_revision_number":6},{"id":"46ab6f9ca987d7a601f34dcf90d28799c7d4f95b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 20:35:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9a9000ab79d4bd5b0e50d074e12bfca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e640410dcfc40dd80fada458e93169b : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c2919727d44472fbb4b55b0af12521b : SUCCESS in 11m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/307faaaeeea441b3b7002e8412299b9d : SUCCESS in 16m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e64dc47e4090498cbb666904811ba862 : SUCCESS in 12m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fac4538021934a368086ae40ed801f20 : SUCCESS in 12m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/d65686b7c41f42eabaf9bbe64993a8bf : SUCCESS in 44m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1004257045ac47dbb1100cb004484702 : SUCCESS in 1h 10m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/043699b90ff845658368ce1a83033f9a : SUCCESS in 47m 39s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/52cb4bf2c4a64bc8b72ee255fc60b5ae : SUCCESS in 17m 45s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/c0fdc4f72fe744eb816faf40f6ba81b5 : SUCCESS in 5m 43s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/83b35a69fba346a98aec983c7c5d395b : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/faf79c96c0d9417992eee2a363104d5c : SUCCESS in 6m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b09fb8383fef4daeaf7ccee7559d5530 : SUCCESS in 6m 11s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/dad379ef1c6b408aa37fa3e9ece31da6 : SUCCESS in 35m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7915d12f8b234b519486f84e4b43216f : SUCCESS in 6m 04s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e12bb0c34b7646428cc065c9ae92038f : SUCCESS in 1h 51m 38s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/9a18f6c79ea446d98d916872551f91f3 : FAILURE in 1h 58m 12s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/489111e75c404b0f83a11ef1a19814ee : FAILURE in 1h 56m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bbcfca44f5374c3c8ccc907e39c12489 : SUCCESS in 1h 43m 41s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/35ea0a13353249888484545eb6f34dce : SUCCESS in 57m 40s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/ac2622d24cbb45b9a64b609e798595f4 : SUCCESS in 1h 47m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/8b3615a36f11439bb82327ccd7dc0b82 : SUCCESS in 35m 44s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0b945a3dff2848a49f739a77ed57038e : FAILURE in 46m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/b5c2f32f6d9f406daf31dda5a55220e9 : SUCCESS in 1h 14m 06s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/94d033ae4931407da8706521691f5b31 : SUCCESS in 2h 36m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/461dbdd9a28c44499919c84fb385df15 : FAILURE in 45m 55s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/207d8fbe94484db28ef13810b7c158c0 : SUCCESS in 10m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"98aaa3ce06840742924b4c34c39e2661da302cb9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-15 20:46:51.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/32/881032/6/check/iscsi-driver-nimble/2848393/2848393.html : FAILURE in 1h 11m 53s","accounts_in_message":[],"_revision_number":6},{"id":"658ee8c229665acdf523cf2d1e5b5f1b34db04d5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-15 22:28:47.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-881032-6/ns5_nfs : SUCCESS in 4h 15s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-881032-6/ns5_iscsi : FAILURE in 4h 4m 21s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"c92ee12cde7f518dc33115cb653bd3b04e16dbe9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-15 23:45:37.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/881032/6/check/DellEMC_Unity_FC/1a6eff9/DellEMC_Unity_FC/276 : SUCCESS in 1h 49m 42s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/881032/6/check/DellEMC_Unity_iSCSI/40f6b3c/DellEMC_Unity_iSCSI/276 : SUCCESS in 1h 52m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"29a1808ab63efab7948a2c298a56572e3ee5b10e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-16 00:28:08.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-881032-6/FC/ : SUCCESS in 4h 14m 17s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-881032-6/iSCSI/ : FAILURE in 5h 45m 11s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"9ef6f538e756233f18b69ab951bf0d5ca23a6e31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-06-16 08:52:40.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"34ce1bec619ca04c91eb12c52e8e879dc1b10b41","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-16 08:53:44.000000000","message":"Patch Set 7:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":7},{"id":"3416c6e1222313fecede86da2c804d9d0822a5f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 09:16:58.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/467288328901431e9905beeb21304ee7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f94f304de45a4297b02e9a54fe29220c : SUCCESS in 13m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3fcb726b207648ce864bec0c8e84f702 : SUCCESS in 21m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ce68073d37a44ee29c664151eb06f778 : SUCCESS in 20m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d37ab3201996318ca0d11f9ad6c60085515c086f","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-16 09:56:26.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/32/881032/7/check/iscsi-driver-alletra-6k/d21cd71/d21cd71.html : FAILURE in 1h 02m 29s","accounts_in_message":[],"_revision_number":7},{"id":"afbd54880bb1d55c7f14a25bd0db17a2d6512c02","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-16 10:15:19.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/881032 : FAILURE in 1h 19m 40s","accounts_in_message":[],"_revision_number":7},{"id":"a32467cf0535e155305b473f6ecfeb002aa32da1","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-16 10:17:06.000000000","message":"Patch Set 7:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/53a908ca5b024d078b16ba1b58cce92a\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/95973c3f7e3a4d26ac8d3771e817d8e0 : SUCCESS in 1h 21m 09s","accounts_in_message":[],"_revision_number":7},{"id":"a322a828d3e8c1fe06c1084b6c0798b9bacdcd74","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-16 10:20:15.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-881032-7/ns5_nfs : FAILURE in 1h 27m 4s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-881032-7/ns5_iscsi : FAILURE in 1h 26m 59s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"63d12809c1a7229a3c75739a6b223943973aa093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 11:16:38.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe26d07e01f04ff2ad9c170c4af14b6a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4337e11c66e843deb883f4c8fe8b7c71 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f30823682a046689f645f54ce524cb5 : SUCCESS in 12m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/452d545dadc349d28b3775d3905a3534 : SUCCESS in 11m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b8c218aee964d7f91d31a10e5c637c2 : SUCCESS in 11m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bf3e306ed1641f0be3630e87a516394 : SUCCESS in 12m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/9adce252c82e4831a3ce387f0f96ae24 : FAILURE in 57m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/60e22bd55684469f8456a01607ac9bf4 : SUCCESS in 1h 38m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d30013198c44dbabf98d23da3ea5865 : SUCCESS in 55m 40s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/2720e004daf74860869c78aaecd432df : SUCCESS in 15m 19s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/5785f0713f4744e7b776a2cf47270ccc : SUCCESS in 4m 54s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6fe24bc11d9246bf957cd1c69f045a2a : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bf2547d648194e71a1208fb8d5ad675a : SUCCESS in 6m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0bb2cf18ee3a481597ad3eb697afb49e : SUCCESS in 5m 50s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/f7ab09642e234904824672e14b0db685 : SUCCESS in 49m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ee8a4469276b469d8afe454a38d1714b : SUCCESS in 6m 10s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1633e7ba636a4e4fbe4ca5b07990c4e2 : SUCCESS in 1h 55m 32s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/96cfc44a234a4ad6b22a3964e8feb614 : FAILURE in 1h 52m 51s (non-voting)\n- cinder-tempest-ceph-multibackend-cbak https://zuul.opendev.org/t/openstack/build/ccb15cbe853148b885094c78ddd8be69 : FAILURE in 2h 10m 19s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3edea933b5264f87b809c02111234afc : SUCCESS in 1h 53m 56s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c873ed24be6a4c0e8927c7244c5e6c84 : SUCCESS in 1h 52m 01s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/2192b12397c84e90a6f0c8b7aa198602 : SUCCESS in 1h 05m 20s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/e8bc143244034c8b9811e6cba1f15682 : SUCCESS in 1h 53m 30s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/2a99bcd144b04de680f6cfe8be44fc40 : SUCCESS in 45m 41s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c3af526fa64b4969aa3bf290c44558e7 : SUCCESS in 1h 31m 13s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e157e9b734e249cab5aab6a7aa01a14a : SUCCESS in 1h 34m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/680af6953b84433f806de6df8533d56c : SUCCESS in 2h 22m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ecaceaa854e4850b76b403246dfdeb1 : SUCCESS in 50m 57s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/15ddfdd6c2654c20a92b894d2982d33e : SUCCESS in 8m 02s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"db644c62c49e215cd1bd5b6394832ed3f14eccfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-06-16 12:12:31.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"07a9bba833769977fe73636915809ddbe0337807","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-16 12:13:31.000000000","message":"Patch Set 8:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":8},{"id":"76c5ac324c1c49ca6e3130b6aee0a5a8a285df62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 12:38:11.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/352dc8922d4943d3891f24204ed77be0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/519364cc7fe145e4bb682393107ddf23 : SUCCESS in 12m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/28398d31fcaa4f5c925b9e83a19456fb : SUCCESS in 23m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b706d39977e1431088df6abdb813619d : SUCCESS in 22m 06s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fc05415b0e97432b892b9542d862baa98abfbc51","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-16 13:12:41.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/881032 : FAILURE in 57m 26s","accounts_in_message":[],"_revision_number":8},{"id":"534d79f83d040503426d0f21e05996d4f51783e0","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-16 13:36:53.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/32/881032/8/check/iscsi-driver-alletra-6k/bcb8fa5/bcb8fa5.html : FAILURE in 1h 21m 37s","accounts_in_message":[],"_revision_number":8},{"id":"64a23985265900a8da65d355074f3478e1f3bf94","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-16 13:40:12.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-881032-8/ns5_nfs : FAILURE in 1h 27m 9s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-881032-8/ns5_iscsi : FAILURE in 1h 27m 8s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"4d5d8459e80d6b2c9afce850ee3cf43e0956868f","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-16 14:02:22.000000000","message":"Patch Set 8:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/9bfdb60f3a2c478f9d610d906ef909a1\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/33d46ec87d0c4909a34db439cf7dc32d : FAILURE in 55m 40s","accounts_in_message":[],"_revision_number":8},{"id":"7ef9fa3dc891dbad55621033ba01d45a0499d5fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 15:31:24.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c116dfc03847467d8d6119999b582dc0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8c5215809e64450a5a5423609e5e257 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bad0f1dca91e4859a01eee93855d9300 : SUCCESS in 11m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c1874b84bc44cafb46a61eb7c3d4361 : SUCCESS in 9m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d38e1770d1c24ff684bd9b6fdb7d13a3 : SUCCESS in 9m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34cececd7c5a46478a59192124fa06d1 : SUCCESS in 11m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/d825b054c3e444d184a26d5c0dd8fc98 : SUCCESS in 49m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a8385f25dee44eb5a6514f9e8a363d70 : SUCCESS in 1h 33m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a82510e0d0c24c35bb1ea097f5a5b161 : SUCCESS in 51m 20s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/1e07da50de7e485787acf431b331222e : SUCCESS in 13m 51s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/e7323503816f4c8580906135e4b6d3a8 : SUCCESS in 6m 00s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/20821c5d3ff14db08d3ffa27227587db : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f4efad165d6e4f5f877c6b6bccd928dc : SUCCESS in 6m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7230dffb30934d78a7ddf03d2b0e0a72 : SUCCESS in 5m 32s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/51a6ee28ecd3405c8c732289237278f7 : SUCCESS in 36m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b72e4f7a42ae425fa1837dfe81e6a679 : SUCCESS in 6m 06s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4d46b3510cf04963ad689c4231f6adfc : TIMED_OUT in 3h 12m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/3028112ff4b142a7a476fc843debb492 : FAILURE in 2h 49m 42s (non-voting)\n- cinder-tempest-ceph-multibackend-cbak https://zuul.opendev.org/t/openstack/build/a661310c651a419d8bb36014cf41c3ad : FAILURE in 1h 57m 56s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/9ca366ac25074f848eb47a0160fca50d : SUCCESS in 1h 48m 34s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/4660aff57f41415db5c89ef46d6ad7dc : SUCCESS in 2h 39m 25s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/392b98000889493f8afda8e91a0e4021 : FAILURE in 1h 33m 00s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/13a20d66753e43f0ac46a2499d4a0eba : SUCCESS in 1h 34m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/83d339a6160040899b78c8a9bcfb6ae2 : FAILURE in 1h 32m 39s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/d19acbe814b641f2a52646fde4d1851e : SUCCESS in 1h 32m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/00510f2854374e83846a371dcf2afdcf : POST_FAILURE in 2h 15m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/de2a57a0b6394846b99f46bbad8eb676 : SUCCESS in 2h 33m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0fcb1a4dfbd4e1fbb1a34befb574934 : SUCCESS in 54m 19s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/cd266b38cd754f188cfb0b0ccbd2e245 : SUCCESS in 10m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9a9720b9ac7da92903c27b3d40aaa23324e14b83","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-16 15:47:34.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/32/881032/8/check/iscsi-driver-nimble/02b137d/02b137d.html : FAILURE in 1h 00m 33s","accounts_in_message":[],"_revision_number":8},{"id":"bdddeec86cb24dc38afd88855da72ec667458c3e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-16 17:25:56.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-881032-8/FC/ : SUCCESS in 2h 44m 19s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-881032-8/iSCSI/ : SUCCESS in 2h 26m 30s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"d4b0c853cd6f3e09be869c28c8e5d4a0c359ca74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-06-16 18:31:57.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"748bf4997f57d5db69b62831a6be63f01f82fbcd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-16 18:32:55.000000000","message":"Patch Set 9:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":9},{"id":"3d7c2c4a6526463518d925f55a362212eede4fa5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 18:45:53.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85d60a0ceebb4f429c080ecde8691c5f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b7070cce3ea9427ea817e3fea4af86ae : SUCCESS in 9m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/83e00a4b828848af807eb35977129563 : SUCCESS in 10m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1b2f70c4231a4336bfcf8afa1792a032 : SUCCESS in 12m 54s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c96690865f11d3ecf3636a750d94d0247c331c8a","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-16 19:48:28.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/881032 : FAILURE in 1h 14m 38s","accounts_in_message":[],"_revision_number":9},{"id":"f58ff222fa82da5e8c4e08db48cff858b09c09a8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-16 19:48:58.000000000","message":"Patch Set 9:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/793bf08094194d86b66e72011647ef5f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/571e7b61d0744224b0ea533c076745ce : SUCCESS in 1h 13m 59s","accounts_in_message":[],"_revision_number":9},{"id":"6fbcd1b494e6c0db02fd6aca37f6b5c15ce91af1","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-16 19:52:39.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/32/881032/9/check/iscsi-driver-alletra-6k/9f46dbd/9f46dbd.html : FAILURE in 1h 19m 25s","accounts_in_message":[],"_revision_number":9},{"id":"b01fe6c428e6551c825fdd07d6931969f83e5fd6","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-16 19:55:20.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/32/881032/9/check/iscsi-driver-nimble/33fb0dc/33fb0dc.html : FAILURE in 1h 21m 39s","accounts_in_message":[],"_revision_number":9},{"id":"dfed9aed279fd60652316799d386346cb1b32e57","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-16 21:16:26.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-881032-9/FC/ : SUCCESS in 2h 18m 42s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-881032-9/iSCSI/ : SUCCESS in 2h 44m 8s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"6a683a5f2c64cc3bc1c18737ca523ad9ef00851f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 21:43:35.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3287cfb2570c4d2381dcb99a8f166264\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3ed2e12c51b40d18c7cebb9a5255c6c : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90af0a17aee34a66a2de8fabe87eeae1 : SUCCESS in 12m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9b42cf555014e33a347eec14e8726dd : SUCCESS in 11m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54e69900b22d4776b861b934a311ec50 : SUCCESS in 10m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e05a8e4777bc42aabe7787a6ffc7b125 : SUCCESS in 15m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/c2e24d537f534996aa105e0b316821ce : FAILURE in 27m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/11207a74abd245609f29e538ccf35758 : SUCCESS in 1h 18m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49d95dbf2ec74b28b3f6a14e3beccfbb : SUCCESS in 49m 48s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/af12cd74025a45f7a1b65ce3e31dd576 : SUCCESS in 16m 34s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/1552449b56dd4e9e8e07bd7f7b9be56e : SUCCESS in 5m 32s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/86b890ac2d6d4b8696e5f4ef9593f98a : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/591817046454420eb5f4cfdffc49cb27 : SUCCESS in 7m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d14cb106c72648bb83f6f94b88854de7 : SUCCESS in 4m 56s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/3c4a4edf3a6e4a8b9accc29114465017 : SUCCESS in 42m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b5f7e7709fe346dcabc2b53332b91eba : SUCCESS in 5m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/b0a732b24c724c0cb5f46400b51e7f28 : SUCCESS in 1h 57m 05s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/0f631c736d8f43dbb7af5fa6ffbe09c3 : FAILURE in 1h 49m 07s (non-voting)\n- cinder-tempest-ceph-multibackend-cbak https://zuul.opendev.org/t/openstack/build/07eb11c68f154938bb96b1a1a16786e0 : FAILURE in 2h 02m 54s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ee8434dcf595406db8732073faa61257 : SUCCESS in 2h 01m 57s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/bed03bdd0e7246338ecb41a970cbb8e7 : SUCCESS in 2h 24m 56s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f857e6022de4420ca680844406264778 : FAILURE in 58m 44s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b13e1ce5a3dc4660b32d10961a6536f6 : SUCCESS in 1h 22m 42s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/06b7ea32870543e384f9ab24cef0c5be : SUCCESS in 46m 16s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/0ab3f888d9344fb8afd52e931d008375 : SUCCESS in 1h 27m 15s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/4c208a4f19df4a8b9acbd9333a985665 : POST_FAILURE in 2h 33m 54s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/79affad4d9004949a1642cd6f00b316d : TIMED_OUT in 3h 04m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18f07f63ddd2433cbb33da0e93781fea : SUCCESS in 1h 02m 49s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/bbcebf55c5cc4e5092a4597ad58ebe5f : SUCCESS in 11m 00s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6869e85426b3058f009cc5ed36cf88a23f709eb7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-16 22:11:14.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/881032/9/check/DellEMC_Unity_FC/e0f3b4e/DellEMC_Unity_FC/291 : SUCCESS in 1h 44m 39s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/881032/9/check/DellEMC_Unity_iSCSI/89a93cc/DellEMC_Unity_iSCSI/293 : SUCCESS in 1h 58m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"f4c6cfe6c3ea7158f135d52ad74972bec20a1a6e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-16 22:29:40.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-881032-9/ns5_nfs : SUCCESS in 3h 55m 8s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-881032-9/ns5_iscsi : SUCCESS in 3h 57m 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"aa886e36ef6e1f09dc0edf78987f1c04a45e0b59","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2023-06-20 14:26:58.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f579821312816796777c88e06c21a84ae62bd195","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-06-20 17:51:31.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"6e8633915413672c57f6e6f162def7797b65626a","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-06-20 17:52:22.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"14f76fe4f988f3ff6f9326fdec021f9f883013d1","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-20 17:52:33.000000000","message":"Patch Set 10:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":10},{"id":"1eee5606dd9c9d2ad5d6e09e9fbc8c7d62ec9d9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-06-20 17:53:39.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"daf563471aed71ee8fd05fb07e72fcf9ded54fd0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-20 17:54:33.000000000","message":"Patch Set 11:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":11},{"id":"6acdb57fa3a26d47a10ed0a575aed846cd7f564f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 18:18:22.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a77679a2a244efea7841850cd88d001\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6eb510dd95684bd19c8ddbee8cb64d53 : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e03425f6fbd4a24ab0877491e52d2e1 : SUCCESS in 22m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/314ee401745848a0a6ac3c945be265dc : SUCCESS in 8m 51s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ab49dd2b869f6e4e106c034a830bd19f04b52666","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-20 19:12:36.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/32/881032/11/check/iscsi-driver-alletra-6k/5413185/5413185.html : FAILURE in 1h 14m 16s","accounts_in_message":[],"_revision_number":11},{"id":"88d4cc4856578ed925cf6a3ff5dd217e905a6fc9","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-20 20:05:11.000000000","message":"Patch Set 11:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/8982ec45838541c297f2b7ecc3837a9f\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/a1b63f06cabd4ea8bacade455b135c84 : SUCCESS in 1h 07m 27s","accounts_in_message":[],"_revision_number":11},{"id":"9b8fd7a609631e771ef56dbc5f09909811ee56fc","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-06-20 20:06:17.000000000","message":"Patch Set 11:\n\nBuild succeeded\n  - For information of Fujitsu ETERNUS CI, see:\n    https://wiki.openstack.org/wiki/ThirdPartySystems/Fujitsu_ETERNUS_CI\n\n- fujitsu-eternus-dx-iscsi http://openstackci.jp.fujitsu.com/Eternusci/32/881032/11/check/fujitsu-eternus-dx-iscsi/41103c4 : SUCCESS in 2h 02m 56s\n- fujitsu-eternus-dx-fc http://openstackci.jp.fujitsu.com/Eternusci/32/881032/11/check/fujitsu-eternus-dx-fc/0cbeba6 : SUCCESS in 2h 09m 53s","accounts_in_message":[],"_revision_number":11},{"id":"940fb303d4998a81ea0114c63eeff0956c15ee28","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-20 20:09:08.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/32/881032/11/check/iscsi-driver-nimble/109f22f/109f22f.html : FAILURE in 1h 08m 30s","accounts_in_message":[],"_revision_number":11},{"id":"79cf7716583b94ea4e91b9dd802fe3826540c872","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-20 20:26:00.000000000","message":"Patch Set 11:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-881032-11/FC/ : SUCCESS in 2h 19m 6s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-881032-11/iSCSI/ : SUCCESS in 1h 50m 11s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":11},{"id":"6e4865050ac5c62395aaf00dc668963843e9d7a3","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-20 20:47:18.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/881032 : FAILURE in 1h 48m 01s","accounts_in_message":[],"_revision_number":11},{"id":"155a93b45c95a61c08e17cfc5cd2e75a237ddbc3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-20 21:21:34.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/32/881032/11/check/DellEMC_PowerFlex/f24bfcf/DellEMC_PowerFlex/170 : SUCCESS in 1h 28m 57s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"f4725c2444f48cd6cb2c7b16819709e72f8e3579","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-20 21:39:59.000000000","message":"Patch Set 10:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-881032-10/FC/ : SUCCESS in 2h 31m 36s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-881032-10/iSCSI/ : FAILURE in 3h 48m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":10},{"id":"2c4817e4c9f383be4f238bc21d6bf05d2991d288","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-20 21:58:58.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-881032-10/ns5_nfs : SUCCESS in 4h 4m 38s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-881032-10/ns5_iscsi : SUCCESS in 4h 6m 57s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"d34ea63ba30c5108bc872d6e9d9aec9fb91cc640","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-20 22:19:01.000000000","message":"Patch Set 11:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-881032-11/ns5_nfs : SUCCESS in 4h 4m 27s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-881032-11/ns5_iscsi : SUCCESS in 4h 3m 29s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":11},{"id":"051318b5a931130bd293fdf484b44094413fb532","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-20 22:35:32.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/881032/11/check/DellEMC_Unity_FC/bd72f3c/DellEMC_Unity_FC/315 : SUCCESS in 1h 52m 50s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/881032/11/check/DellEMC_Unity_iSCSI/3ac73cb/DellEMC_Unity_iSCSI/316 : FAILURE in 1h 49m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"c95ac8d84b7e9f991e8665fa086e00d271cf0162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 00:33:43.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98e6415550d9400e91c12715282d69c4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed5e6d6ef7d0441390cb74a22f097a7d : SUCCESS in 8m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cf03d819be0417cb1abf83907af0d90 : SUCCESS in 15m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80154bb8d5404cb6835e0f44c1e1e90f : SUCCESS in 12m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/76ba78627ddb4cd09f6223bb0f10c53f : SUCCESS in 12m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/618a8c21216a4676ac6efcf9bf6cd7a7 : SUCCESS in 11m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/301fa61c3ebb461eb07dd6f1ff4c2f01 : SUCCESS in 40m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/305d92e6c49e454f8dc91bf54159bc33 : SUCCESS in 1h 26m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eae1cd2163fe4214b11f9c9573aee389 : SUCCESS in 52m 15s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/f37086d971574d98852f413f11b6d97d : SUCCESS in 15m 08s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/9257e5baeceb4768aee5fbeb4c876bc9 : SUCCESS in 5m 16s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b06dc08097f840f6b3719c7128f116f8 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4cf4c1cd5cd44b538166f5497de23cd9 : SUCCESS in 6m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/add3f95c45e24f83a816858a50cb84b5 : SUCCESS in 6m 14s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/bbfbef312bb94db9b66564079dcd4a48 : SUCCESS in 37m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cf151a3e4cb343b198ddd03a41ebe94a : SUCCESS in 6m 58s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4e34e632cb73469fbce0897d41f25d64 : SUCCESS in 2h 36m 41s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d32b67852caa4e4997a691481c0e7643 : FAILURE in 2h 13m 09s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/40627f3bf7c24b99a9ad0d72c80e5e12 : FAILURE in 2h 33m 32s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/c4233e5b470d4144abdde471858048e8 : SUCCESS in 2h 30m 55s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/1dee62b8c59f4d4788bd51357b8b54f5 : SUCCESS in 1h 16m 49s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/86b6af29f8ad438fb8715f8670a4c928 : SUCCESS in 2h 06m 44s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/71cf6483e4b94cd0ba93aa7056508d89 : FAILURE in 1h 00m 28s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/f421b48646d5494baf1f1309b7e9f163 : SUCCESS in 1h 22m 00s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/e5c0ae6205d243fb97c7baed73dd27b4 : SUCCESS in 1h 45m 08s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e7fd41eccc9f485ba9a43593226bd968 : SUCCESS in 2h 23m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b001d8a44f91475fa10bd51ae0b89b6b : SUCCESS in 57m 45s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/a4fe5468c150425c8a6c0e66fc59b38c : SUCCESS in 12m 00s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7393352da8582261ec7a93f659679f60c5028931","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-21 04:12:04.000000000","message":"Patch Set 6:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/32/881032/6/fc : SUCCESS in 0h 29m 11s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/32/881032/6/iscsi : SUCCESS in 0h 40m 25s","accounts_in_message":[],"_revision_number":6},{"id":"1889430c707324c1b70c90d6e1596f1ea664ad85","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-21 04:12:08.000000000","message":"Patch Set 7:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/32/881032/7/fc : SUCCESS in 0h 33m 13s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/32/881032/7/iscsi : SUCCESS in 0h 40m 56s","accounts_in_message":[],"_revision_number":7},{"id":"25d488837c26c4baf2a830ff877966cc27359904","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-21 04:12:11.000000000","message":"Patch Set 8:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/32/881032/8/fc : SUCCESS in 0h 50m 19s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/32/881032/8/iscsi : SUCCESS in 1h 03m 19s","accounts_in_message":[],"_revision_number":8},{"id":"b1f23afaf8c76d24947a713f70979106ea33969a","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-21 04:12:15.000000000","message":"Patch Set 9:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/32/881032/9/fc : SUCCESS in 0h 40m 05s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/32/881032/9/iscsi : SUCCESS in 0h 43m 44s","accounts_in_message":[],"_revision_number":9},{"id":"6b3dc5ccb0fbafe29186b42893f09c5e280a520e","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-06-21 09:07:42.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"ad484b88be307e8f07cdbbb0f2a93ca7a0ccb967","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-06-21 09:07:48.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"eeb8cde7e0f1aaaac014059713fd2966fc53ddb0","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-06-21 09:08:26.000000000","message":"Patch Set 11:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":11},{"id":"165d27b966ef1df78bbbe1a5f47f4212623621b0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-21 09:08:55.000000000","message":"Patch Set 11:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":11},{"id":"27d42a431ee478d628e5b621b5ff8459e0553b31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 09:22:49.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae64f5ce33784fe2b63d78e8841c0570\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3704bd3f2064cc298d67029887873d5 : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5efc5efd95e4a5aa43d437c8d12b444 : SUCCESS in 10m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0a88b23e1ca24fe4ac0e1bf445bc49cb : SUCCESS in 10m 16s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a50625bda5815599d1e56d1fb6aeff5fbf5e9287","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-21 11:35:33.000000000","message":"Patch Set 11:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/9e3824a5e9b444eaab919314697ba54e\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/89e5b28a2ab8433f97b92a956d189a25 : SUCCESS in 1h 13m 25s","accounts_in_message":[],"_revision_number":11},{"id":"0ca66974811aa304ff5122fd20f0a8b29bc6ebf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 14:24:04.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed35f7c5f7af40ca81307f26e668dc70\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f69b94f98174924af2848f6b2f07e9f : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82b09c59b0bc4464814d4cf3ee144d8d : SUCCESS in 14m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a26b1701475f44749520c1cf4c0e4be1 : SUCCESS in 13m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/634baf4e4075486897136f48ecfb8632 : SUCCESS in 11m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6de34aefb467468cb3c46a0cfed6b79f : SUCCESS in 13m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/ffaaf3449f4448d9b2884f884428a3d3 : SUCCESS in 1h 06m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/86ef53cfa75840628c8beecf43d97740 : FAILURE in 1h 53m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ba69b094a4b4c31a224ab5140a04794 : SUCCESS in 59m 52s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/6038cad9c9cd44a3b176ab4c644ad116 : SUCCESS in 13m 05s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/b5dc977f30db49e8b7fac3da0f14dcc3 : SUCCESS in 6m 23s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/adf3747c545b4a7cb2fa0e5eae8e472a : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1e8f6e72004744598cbcb5653374939e : SUCCESS in 5m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2f4c8e1f2c314e4e93db109efd687bbd : SUCCESS in 5m 51s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/4a7cb81314254856b1d9296f691b58ce : SUCCESS in 1h 09m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ac2655bc73eb437b809db7a2fa3a9ea0 : SUCCESS in 5m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/068348546d87435ebf70cbc076fe6375 : SUCCESS in 1h 53m 14s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/2f9f382c7a134d0eac828008e1ec1d15 : FAILURE in 2h 12m 27s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6b6cc58c720a4d049b5925ae45d26943 : SUCCESS in 2h 17m 27s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/b1c1a612cf9f4ed7865566f287dcabc4 : SUCCESS in 2h 08m 49s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/459d6c704a4f42c1ac217c83f088b0c9 : SUCCESS in 1h 29m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/9e792486e9124b31a0a3f73b6d0b2f73 : SUCCESS in 1h 46m 18s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/6d3cf152c35e4f95aab0e89bcde770f1 : SUCCESS in 34m 46s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/da568e1a2dec470e9ffd44faf7dd7095 : SUCCESS in 1h 28m 14s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/28a3ce1366b048728a366917e7cf9cc0 : POST_FAILURE in 2h 55m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9e6b162380674d11b790da91329f8fe1 : SUCCESS in 2h 23m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e926aa9a98d94c00997285b242f8b3c5 : SUCCESS in 1h 00m 44s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/5c6e7f57791840ca831940f835493e99 : SUCCESS in 11m 18s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"2e459faeb7c0c7151238443d00c1fa13cf4ecffa","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-06-22 15:16:14.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"58d899c83e8df2ad8b8cbef24b063114ec6c94c3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-06-23 13:28:46.000000000","message":"Topic set to rbd-backup-restore","accounts_in_message":[],"_revision_number":11},{"id":"eb3f93cca86d499496c911b52db152ad1b145b3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-06-23 14:48:16.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"c3449bf836ddaaf27e973b2f99c7bed987bd25fc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-23 14:49:09.000000000","message":"Patch Set 12:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":12},{"id":"b1361c815b2d598f954d2a7c437ecb0d414014e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 14:59:55.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13837c4ea2024017886a3815e0ff0f25\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ec3f19ca8ad40ccba194f8bb5befec5 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f4dfc9898d394da48c01b1e3df50de23 : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5c3998111c7a454c8c45fcca3598577a : SUCCESS in 10m 34s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"0ed82375bdd3801104df5fc1f307dcc28d0b3b7a","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-06-23 15:48:22.000000000","message":"Patch Set 11:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/881032/11/20230623184716/NEC-V-FC : SUCCESS in 3h 26m 44s\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/881032/11/20230623184716/NEC-V-ISCSI : FAILURE in 6h 01m 03s","accounts_in_message":[],"_revision_number":11},{"id":"364c1bffd6c5ccb59cf9e854b228ff79f2a5f6fc","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-23 16:03:54.000000000","message":"Patch Set 12:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/e70bda59378a42c9be6ca48886593700\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/22e1f7af1f904b3d90054b9aa7632ddd : SUCCESS in 1h 11m 57s","accounts_in_message":[],"_revision_number":12},{"id":"2da059434146659b219bb37bf630df34fadf6d5c","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-23 16:08:21.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/32/881032/12/check/iscsi-driver-alletra-6k/8741586/8741586.html : FAILURE in 1h 18m 24s","accounts_in_message":[],"_revision_number":12},{"id":"5ed974fc259ffbc2b14b40c45cbe04be03a7d85b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-06-23 16:10:49.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/881032 : FAILURE in 1h 19m 23s","accounts_in_message":[],"_revision_number":12},{"id":"1c06651e2498eb0fcf93ad71cc26c9ab51478a9e","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-23 16:49:45.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/32/881032/12/check/iscsi-driver-nimble/2c4bc0c/2c4bc0c.html : FAILURE in 1h 17m 51s","accounts_in_message":[],"_revision_number":12},{"id":"18440f95e836d3923c0d6b90f7e27139211fd406","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 17:16:56.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2e57e4b6f9e48c6a6873cc98c7df715\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/993d2dc226664838905db4ae6a28a5e5 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fd65a0a90d945b1af373b7dcb1bfc1c : SUCCESS in 10m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32ccc89abdde413bb5d8f330d6768d43 : SUCCESS in 15m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb664e1eccbf4ad5a2c00c1f0eb002ba : SUCCESS in 12m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20c8533d38c24574a686fb95a654cc41 : SUCCESS in 14m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/7bcad85bf190429f87f17f22d23264ff : SUCCESS in 57m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5f42f4f0576d4d1d8d81fac71f4da4f8 : FAILURE in 56m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/823fe34c55af4a779c788dad05a9e702 : SUCCESS in 1h 01m 51s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/ca9e3a2acedf462b88f3e59fb65d5e22 : SUCCESS in 15m 17s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/d7e4ef5e6cf04559af492921a25a5ff6 : SUCCESS in 6m 03s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f9a2823e1383468da8087c31835f520f : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f6f5ef0483a448d99989b9a8ff91978d : SUCCESS in 7m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2f795cc151fd4f689d06885d50f12366 : SUCCESS in 7m 20s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/8a426fa456244b56b1205f18df5fba1e : SUCCESS in 42m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/54f0a751701447c681e15237ff21ba29 : SUCCESS in 6m 45s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3dd282be07a141c7a5175eafbee8598f : SUCCESS in 2h 24m 52s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/d90c07e6875b4cfd85ab31818c790bae : FAILURE in 2h 09m 47s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f96036ea239b4f4cbe857f9ba8e093f7 : SUCCESS in 2h 27m 04s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/fc8bef57a5bc4ef7bfcb05db29caac9a : FAILURE in 2h 13m 32s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/831833ff462945e6b7123c71c5242c5c : SUCCESS in 1h 09m 23s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0db99a2fb14a48efb5649a3460d85392 : FAILURE in 2h 14m 59s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/b0c2d217390a43e78603522c5cb490c5 : TIMED_OUT in 2h 08m 40s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c72cbd608d664ba1b8d53bdf1a6e123e : SUCCESS in 1h 39m 02s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/5dc3b7c1b3224d07a414da0039dc903e : FAILURE in 2h 09m 42s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4de93b15cd2a480ca6fd6d61fe3c0d02 : FAILURE in 1h 20m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13713ca87ddd47c3b64b82642917eced : SUCCESS in 41m 45s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/5d700ab3a0cc4d5fa353d7b964405552 : SUCCESS in 14m 14s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"da063a40debf95bf9ce245402e9985ff88b3c5ad","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-23 18:50:41.000000000","message":"Patch Set 12:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-881032-12/ns5_nfs : SUCCESS in 4h 1m 54s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-881032-12/ns5_iscsi : SUCCESS in 4h 1m 37s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":12},{"id":"8744d044df8221a1372559285cb43eb4e7d04bd2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-23 19:25:18.000000000","message":"Patch Set 12:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-881032-12/FC/ : FAILURE in 4h 36m 39s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-881032-12/iSCSI/ : SUCCESS in 2h 4m 37s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":12},{"id":"45d206213b1f8b3c7333d5312e8204f58602f30f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-23 20:12:58.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/32/881032/12/check/DellEMC_PowerFlex/3392693/DellEMC_PowerFlex/197 : FAILURE in 1h 31m 44s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"a04da226fbacfce4b16cf3e642561e32f73ebfc7","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-23 22:00:07.000000000","message":"Patch Set 11:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/32/881032/11/fc : SUCCESS in 0h 26m 24s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/32/881032/11/iscsi : SUCCESS in 0h 41m 24s","accounts_in_message":[],"_revision_number":11},{"id":"d5e47ed05169f601ac394260d34fd9d1781f29c9","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-25 21:00:07.000000000","message":"Patch Set 12:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/32/881032/12/fc : SUCCESS in 0h 34m 17s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/32/881032/12/iscsi : SUCCESS in 0h 45m 18s","accounts_in_message":[],"_revision_number":12},{"id":"4530462a8cb6686191329c3c0aab0f3ce8717c8c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-26 22:35:33.000000000","message":"Patch Set 12:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-881032-12/FC/ : UNSTABLE in 6h 57s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-881032-12/iSCSI/ : FAILURE in 10h 4m 47s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":12},{"id":"82611b00b4957837e0c4365377e36995f8b0db62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-06-27 13:35:13.000000000","message":"Patch Set 13: Published edit on patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"a1dd1859cb7a4c1b2dfcafbf27d54546d9f4c72f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-27 13:36:04.000000000","message":"Patch Set 13:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":13},{"id":"4510a0eab78e84e84bc7b5dc16c7c4d138ae6a70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 13:51:55.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a0ab4a637734a94bab847a41dd90a4b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d7cd8c843ca473fb54483e041a812e9 : SUCCESS in 10m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b4af7a9ff3f74c32a65dcda8ac02448a : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3256a6362de34d6b8dc899b632dced57 : SUCCESS in 11m 05s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"c8fd737bfa2b119e2c198b9347240420a74ac9fc","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-06-27 14:05:57.000000000","message":"Patch Set 13:\n\nBuild failed.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/32/881032/13/check/tyds_test/d5ac23b : FAILURE in 30m 15s","accounts_in_message":[],"_revision_number":13},{"id":"e5224f6109c710f73a54eadf211cab92a766da33","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-27 14:50:56.000000000","message":"Patch Set 13:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/b024beb26ce84af2b2acb5c46a093072\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/25f3aa6a554e4fe9a726b17666d95d04 : FAILURE in 1h 13m 07s","accounts_in_message":[],"_revision_number":13},{"id":"82a6d2c3f829bbe3de54654fb0949d57d469a498","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-27 17:12:21.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/32/881032/13/check/iscsi-driver-nimble/8f72198/8f72198.html : FAILURE in 1h 02m 44s","accounts_in_message":[],"_revision_number":13},{"id":"3dd32338f72ee63632ab5f899f1f9dcfddd417f0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-27 17:47:31.000000000","message":"Patch Set 13:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-881032-13/ns5_nfs : SUCCESS in 4h 11m 48s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-881032-13/ns5_iscsi : SUCCESS in 4h 3m 57s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":13},{"id":"6544ed9da0f378e4bbc7eedd75ec0e02ce80c521","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-06-27 18:26:21.000000000","message":"Patch Set 13:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1512/","accounts_in_message":[],"_revision_number":13},{"id":"70dc3f8529542a2cec4c2e34eb81b4cdea063b06","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-27 19:23:43.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/32/881032/13/check/DellEMC_PowerFlex/097b62c/DellEMC_PowerFlex/216 : SUCCESS in 1h 52m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":13},{"id":"b7d83b2d8765a86a9a8b73f683cffb9c6b635054","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-27 20:00:07.000000000","message":"Patch Set 13:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/32/881032/13/fc : SUCCESS in 0h 32m 32s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/32/881032/13/iscsi : SUCCESS in 0h 47m 55s","accounts_in_message":[],"_revision_number":13},{"id":"3cca96b7626b5a729f5bfb7dbac869dd5a76ade8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-27 20:07:30.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/881032/13/check/DellEMC_Unity_FC/907a46b/DellEMC_Unity_FC/337 : FAILURE in 1h 43m 31s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/881032/13/check/DellEMC_Unity_iSCSI/21c2f9c/DellEMC_Unity_iSCSI/336 : FAILURE in 2h 00m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"5b77fc966dc6b47c2742c59d60258c76e0d89859","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 20:51:44.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c44915001bba4b6ca9f3659564260287\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf7d01c08c4b4b7b8635a5e4f35db1a3 : SUCCESS in 9m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/722258648d4242b6b355673140084df5 : SUCCESS in 10m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55684efaf8f14aa1a74f487fb2a0afb5 : SUCCESS in 13m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17823e55ea8b48dc96210200bb9c7429 : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eac816999a034dfa9ca397c76a1c0bcd : SUCCESS in 14m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/909ba149d26a4d6e84c9cd62ac45a664 : SUCCESS in 1h 03m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/52a583be02dc4c9e84b427a4f7b1a785 : SUCCESS in 1h 41m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a1d064e59dd4affaf9d7c54e56efc1c : SUCCESS in 1h 11m 28s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/0358a012994e4ca3b406bd2ad35580b5 : SUCCESS in 22m 49s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/2e603c348bb5435a87bd441f6d4fc682 : SUCCESS in 5m 15s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7d73898a3d7e40deb623366d89911385 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8c6cf54d73f04e749286c87a704069c0 : SUCCESS in 6m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5db258b1e91d4036be07343480b531fc : SUCCESS in 6m 06s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5539ed268f0742b19295cd828626a7cb : SUCCESS in 53m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3656c1e603a54a4a88be770f1fe3d816 : SUCCESS in 6m 21s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5df4af0f61574a64b30405620e018bad : SUCCESS in 2h 10m 51s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/a6ae171f56bd4139b8871b6a9c9a85dd : SUCCESS in 2h 04m 23s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/54630fafe7354e8dae1dd933f65d54be : SUCCESS in 2h 40m 58s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/e9cf719f34174f4cb9973a16115e1a7e : SUCCESS in 2h 41m 50s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/16481562d97946e780d9d0edde2d5f4e : SUCCESS in 1h 34m 31s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/195aa372b9ca4b938565ceb3662cb5a4 : SUCCESS in 2h 22m 51s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e7dd084ad1c34014b00088a5a78efb5f : SUCCESS in 35m 58s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/702f17a37d0d4466a10f66ac1a9227d9 : SUCCESS in 1h 43m 07s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/83a9067df09e467885d28c5bf4b07455 : POST_FAILURE in 2h 51m 01s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d505da17cf5f4fafaa06a06f59aacd90 : TIMED_OUT in 3h 03m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80834b35863648e7b22fee7f2bb3516e : SUCCESS in 1h 03m 48s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/f3598a5fb64d47a29705bc5d37ce78a2 : SUCCESS in 12m 11s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"67f7f3335e38f5e6796e323dd46a1a122a4b65ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-06-28 10:39:21.000000000","message":"Patch Set 14: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"7ae6eaeb60a932fcaf6583eafacce8d330034d94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2023-06-28 10:39:27.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"51eedfc355d8c73cd8eb547866dfc27d3bce57e3","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-06-28 10:39:53.000000000","message":"Patch Set 14:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":14},{"id":"4510b71a8e1477a20d521c24257ac18c2179010f","author":{"_account_id":18120,"name":"Fujitsu ETERNUS CI","email":"fj-lsoft-eternusci@dl.jp.fujitsu.com","username":"Fujitsu_ETERNUS_CI","tags":["SERVICE_USER"]},"date":"2023-06-28 10:40:00.000000000","message":"Patch Set 15:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":15},{"id":"9813ac038b63fd59ad3d58696af54e16c20d8881","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-06-28 10:40:32.000000000","message":"Patch Set 15:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":15},{"id":"3194ea80f8cac357bbcf1ade833928cc31f55b14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 10:44:29.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42378e83839c4d0fa760be68791ee16d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/edf21d41480c4f4f9c89cdd38816fa4b : NODE_FAILURE Node request 300-0021564089 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e9442091c754d52926650a1ab72fe61 : NODE_FAILURE Node request 300-0021564090 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/29bac5d9fecb4df7a8d6964bbd2fb874 : NODE_FAILURE Node request 300-0021564091 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"954f8ebda7f5766eeaf482776542e1688e9f1c71","author":{"_account_id":32418,"name":"TOYOU TYDS CI","email":"caiql@toyou.com.cn","username":"caiqilong"},"date":"2023-06-28 10:45:55.000000000","message":"Patch Set 15:\n\nBuild failed.  To recheck use \u0027run-TOYOU TYDS CI\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/TOYOU_TYDS_CI\n\n- tyds_test http://119.28.156.143:9090/32/881032/15/check/tyds_test/253581f : FAILURE in 3m 58s","accounts_in_message":[],"_revision_number":15},{"id":"0c53bd363d12661a1235ff13bcfa9115406040d0","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-06-28 11:54:49.000000000","message":"Patch Set 15:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/cd8f9328f6904231a0b2f951120ed9d4\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/df65bfe7252c45d4840c2eaa85e57a62 : SUCCESS in 1h 11m 58s","accounts_in_message":[],"_revision_number":15},{"id":"972e7e352a5e06bbc487cc30f10cbb2fd9e3e4b9","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-06-28 12:26:54.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- iscsi-driver-nimble https://hpe-storage.github.io/nimble-ci-logs/logs/32/881032/15/check/iscsi-driver-nimble/2f69403/2f69403.html : FAILURE in 1h 23m 13s","accounts_in_message":[],"_revision_number":15},{"id":"46feebe0cd5fc74c2420833dc103512aeb5f1df8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-06-28 13:21:45.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/32/881032/15/check/DellEMC_PowerFlex/5243a3a/DellEMC_PowerFlex/224 : FAILURE in 1h 44m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"195ee150a2069e0e3cfd2ceef3af883a33728432","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 13:56:39.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/205236887e694c4da233c930adc60f57\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f5c9d40626940d1a807372b193a1303 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94fd4a541440480ba9312ca062958faa : SUCCESS in 12m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d2c7d0a1d864dc1bfdd8a856bb31499 : SUCCESS in 18m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/412f9264902649c5a896702ee6a06780 : SUCCESS in 14m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21ae2a7b636643a3b7cd5c86e7cf0ae8 : SUCCESS in 18m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/6b159828687e45d898f31089de40bcc3 : SUCCESS in 55m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ce6d1f6c9eb543b2976a1d73f6167be8 : TIMED_OUT in 2h 04m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31f62a98a4cd4941a65c383f53d34fb1 : SUCCESS in 1h 01m 30s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/5deac40e64104f97b09dce25ceabcf0c : SUCCESS in 22m 16s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/963e64a34e3d4dc2b9b50d4f257fd179 : SUCCESS in 5m 21s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ecb5c87ed07a44d4aa66ac5d09516a32 : SUCCESS in 11m 18s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e69f163122584f62b61e892152ace479 : SUCCESS in 9m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f569bd2d52d745c192521a94b343069a : SUCCESS in 8m 57s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/cad86ab744ae404baaea8dec3ee670bc : SUCCESS in 37m 15s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f4e01b9ed8f0487fb3e11d8f09d97831 : SUCCESS in 12m 33s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8587fe9f53364a9f9c680bb4e8f52ef7 : SUCCESS in 1h 40m 10s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/85d5ea2868a44ed09922f1ee6bbef1c3 : FAILURE in 2h 37m 34s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/fdea799e1a694c03ad9c6fc09f9e5298 : SUCCESS in 2h 24m 51s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/8b67f07096fe49918d667f233951f444 : SUCCESS in 2h 05m 12s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/368cc444bcab4e2cbea6c69a07963309 : SUCCESS in 1h 13m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/562fa8e0839c43a4ac740c69adf50c3c : SUCCESS in 1h 48m 58s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/e45b7dbedf4943c1a3397f85f4001cb4 : SUCCESS in 44m 56s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/09220e40087b42f880240d6464144be2 : SUCCESS in 1h 32m 19s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/8e43c122ff654c2e9d673c611a5fdccd : SUCCESS in 1h 20m 15s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d04540ef6017400bb6f4ae0f11496b3d : TIMED_OUT in 3h 07m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/732b1ad27d4c43218096e55a3095fa85 : SUCCESS in 1h 11m 12s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/958ec374b7d64058a2f2d29fb33b3e8b : SUCCESS in 11m 51s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"d1c3c1e270ac16e16ddf45d732f2c950464e12d6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-28 13:58:35.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/32/881032/15/check/DellEMC_Unity_FC/1d1f76e/DellEMC_Unity_FC/346 : ABORTED in 2m 14s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/32/881032/15/check/DellEMC_Unity_iSCSI/7f1a725/DellEMC_Unity_iSCSI/345 : FAILURE in 1h 13m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"82a362ddb1d88a1acaea4f286d25371cdba983be","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-28 14:43:51.000000000","message":"Patch Set 14:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-881032-14/ns5_nfs : SUCCESS in 3h 59m 4s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-881032-14/ns5_iscsi : SUCCESS in 4h 3m 59s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":14},{"id":"c488e6a40f175d97dbb7b43f5ecc10d19c21a45f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-28 15:02:54.000000000","message":"Patch Set 15:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-32-881032-15/ns5_nfs : SUCCESS in 3h 55m 22s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-32-881032-15/ns5_iscsi : SUCCESS in 4h 33s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":15},{"id":"a8dba91850472598e903dc2440c3bb06c611ff60","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-28 16:00:08.000000000","message":"Patch Set 14:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/32/881032/14/fc : SUCCESS in 0h 54m 13s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/32/881032/14/iscsi : SUCCESS in 1h 07m 22s","accounts_in_message":[],"_revision_number":14},{"id":"ae975e1bb5970f6a1f1a656f265a31f666cbb620","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-06-28 17:00:08.000000000","message":"Patch Set 15:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/32/881032/15/fc : SUCCESS in 0h 55m 30s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/32/881032/15/iscsi : SUCCESS in 1h 01m 15s","accounts_in_message":[],"_revision_number":15},{"id":"e36f55b4e4350ccd5a280ab1c85ef1ae36e910fb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-06-28 20:04:52.000000000","message":"Patch Set 15:\n\nBuild success. \nISCSI build logs link: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1519/","accounts_in_message":[],"_revision_number":15},{"id":"b2c594b97969edcce9e9e874bddcb24224961df2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-06-30 22:10:32.000000000","message":"Patch Set 15:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-881032-15/FC/ : SUCCESS in 2h 31m 17s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-881032-15/iSCSI/ : SUCCESS in 2h 1m 27s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":15},{"id":"a966752bded026366b6e4fb130ecef241fd2e9cf","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-07-03 23:38:52.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/881032/15/20230704055223/NEC-V-ISCSI : SUCCESS in 2h 27m 37s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/881032/15/20230704055223/NEC-V-FC : SUCCESS in 2h 46m 25s","accounts_in_message":[],"_revision_number":15},{"id":"121cca6b256f4aa2ec1deb2e8881923e37013e69","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-07-04 02:06:22.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/32/881032/15/20230704071949/NEC-ISCSI : SUCCESS in 3h 27m 37s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/32/881032/15/20230704071949/NEC-FC : SUCCESS in 3h 46m 30s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"77d6d9c6f05c9eff139634334537d9b906eb7097","revisions":{"d62b9c6f7daaddb84e46d254e80ab551537cc387":{"kind":"REWORK","_number":1,"created":"2023-04-20 17:12:24.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/32/881032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/881032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/881032/1"}}},"commit":{"parents":[{"commit":"2e8aff660b18d1e8f961d6de979d541a8d06f1c1","subject":"Merge \"Add fips check jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2e8aff660b18d1e8f961d6de979d541a8d06f1c1"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-04-20 17:10:24.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-04-20 17:10:24.000000000","tz":0},"subject":"Zuul: Job For LVM + Ceph With c-bak Enabled","message":"Zuul: Job For LVM + Ceph With c-bak Enabled\n\nThis patch adds a non-voting job to test backup and multibackend\nfeatures, such as restoring backups from different volume types.\n\nDepends-On: Iff8e1e90ab3c7b519819577ec3aafff838e6934f\nChange-Id: Ibf2b50a23bff926a21c07ba2cc00ed16d956cf54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d62b9c6f7daaddb84e46d254e80ab551537cc387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d62b9c6f7daaddb84e46d254e80ab551537cc387"}]},"branch":"refs/heads/master"},"8d668027b91752bf7af5011e35a0e8a9ab72a615":{"kind":"REWORK","_number":2,"created":"2023-04-21 10:16:27.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/32/881032/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/881032/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/881032/2"}}},"commit":{"parents":[{"commit":"2e8aff660b18d1e8f961d6de979d541a8d06f1c1","subject":"Merge \"Add fips check jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/2e8aff660b18d1e8f961d6de979d541a8d06f1c1"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-04-20 17:10:24.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-04-21 10:16:08.000000000","tz":0},"subject":"Zuul: Job For LVM + Ceph With c-bak Enabled","message":"Zuul: Job For LVM + Ceph With c-bak Enabled\n\nThis patch adds a non-voting job to test backup and multibackend\nfeatures, such as restoring backups from different volume types.\n\nDepends-On: Iff8e1e90ab3c7b519819577ec3aafff838e6934f\nChange-Id: Ibf2b50a23bff926a21c07ba2cc00ed16d956cf54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d668027b91752bf7af5011e35a0e8a9ab72a615"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/8d668027b91752bf7af5011e35a0e8a9ab72a615"}]},"branch":"refs/heads/master"},"b09fb253837cceb2418a386fcb354a94ede9c6bf":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-04-26 11:03:41.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/32/881032/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/881032/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/881032/3"}}},"commit":{"parents":[{"commit":"40147e2489c1f4f9940dc3df2cfa0fa81b5070df","subject":"Merge \"Require novaclient \u003e\u003d 18.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40147e2489c1f4f9940dc3df2cfa0fa81b5070df"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-04-20 17:10:24.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-04-26 11:03:41.000000000","tz":0},"subject":"Zuul: Job For LVM + Ceph With c-bak Enabled","message":"Zuul: Job For LVM + Ceph With c-bak Enabled\n\nThis patch adds a non-voting job to test backup and multibackend\nfeatures, such as restoring backups from different volume types.\n\nDepends-On: Iff8e1e90ab3c7b519819577ec3aafff838e6934f\nChange-Id: Ibf2b50a23bff926a21c07ba2cc00ed16d956cf54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b09fb253837cceb2418a386fcb354a94ede9c6bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b09fb253837cceb2418a386fcb354a94ede9c6bf"}]},"branch":"refs/heads/master","description":"Rebase"},"0a46ebb751d4bd5211c7b4b98814e288abc511af":{"kind":"REWORK","_number":4,"created":"2023-05-15 12:57:41.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/32/881032/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/881032/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/881032/4"}}},"commit":{"parents":[{"commit":"40147e2489c1f4f9940dc3df2cfa0fa81b5070df","subject":"Merge \"Require novaclient \u003e\u003d 18.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40147e2489c1f4f9940dc3df2cfa0fa81b5070df"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-04-20 17:10:24.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-05-15 12:54:58.000000000","tz":0},"subject":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak","message":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak\n\nThis patch adds a periodic job to test restoring backups.\n\nDepends-On: Iff8e1e90ab3c7b519819577ec3aafff838e6934f\nChange-Id: Ibf2b50a23bff926a21c07ba2cc00ed16d956cf54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a46ebb751d4bd5211c7b4b98814e288abc511af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0a46ebb751d4bd5211c7b4b98814e288abc511af"}]},"branch":"refs/heads/master"},"b1e35001c618f7d365b9ebacdfd5b4b47e1b24e1":{"kind":"REWORK","_number":5,"created":"2023-06-15 11:48:42.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/32/881032/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/881032/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/881032/5"}}},"commit":{"parents":[{"commit":"40147e2489c1f4f9940dc3df2cfa0fa81b5070df","subject":"Merge \"Require novaclient \u003e\u003d 18.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40147e2489c1f4f9940dc3df2cfa0fa81b5070df"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-04-20 17:10:24.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-06-15 11:48:32.000000000","tz":0},"subject":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak","message":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak\n\nThis patch adds a periodic job to test restoring backups.\n\nDepends-On: Iff8e1e90ab3c7b519819577ec3aafff838e6934f\nChange-Id: Ibf2b50a23bff926a21c07ba2cc00ed16d956cf54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b1e35001c618f7d365b9ebacdfd5b4b47e1b24e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b1e35001c618f7d365b9ebacdfd5b4b47e1b24e1"}]},"branch":"refs/heads/master"},"d3a210f792aff102a3daa131f3e5793f52f77c00":{"kind":"REWORK","_number":6,"created":"2023-06-15 17:56:40.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/32/881032/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/881032/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/881032/6"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-04-20 17:10:24.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-06-15 17:56:13.000000000","tz":0},"subject":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak","message":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak\n\nThis patch adds a periodic job to test restoring backups.\n\nDepends-On: Iff8e1e90ab3c7b519819577ec3aafff838e6934f\nChange-Id: Ibf2b50a23bff926a21c07ba2cc00ed16d956cf54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d3a210f792aff102a3daa131f3e5793f52f77c00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d3a210f792aff102a3daa131f3e5793f52f77c00"}]},"branch":"refs/heads/master"},"fd7e3948e3103aa169739b4f0ddf2704e936d730":{"kind":"REWORK","_number":7,"created":"2023-06-16 08:52:40.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/32/881032/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/881032/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/881032/7"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-04-20 17:10:24.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-06-16 08:52:06.000000000","tz":0},"subject":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak","message":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak\n\nThis patch adds a periodic job to test restoring backups.\n\nDepends-On: Iff8e1e90ab3c7b519819577ec3aafff838e6934f\nChange-Id: Ibf2b50a23bff926a21c07ba2cc00ed16d956cf54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fd7e3948e3103aa169739b4f0ddf2704e936d730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/fd7e3948e3103aa169739b4f0ddf2704e936d730"}]},"branch":"refs/heads/master"},"53360ea56d4334412edcb1c9d81d029df17bfa41":{"kind":"REWORK","_number":8,"created":"2023-06-16 12:12:31.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/32/881032/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/881032/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/881032/8"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-04-20 17:10:24.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-06-16 12:11:53.000000000","tz":0},"subject":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak","message":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak\n\nThis patch adds a periodic job to test restoring backups.\n\nDepends-On: Iff8e1e90ab3c7b519819577ec3aafff838e6934f\nChange-Id: Ibf2b50a23bff926a21c07ba2cc00ed16d956cf54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/53360ea56d4334412edcb1c9d81d029df17bfa41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/53360ea56d4334412edcb1c9d81d029df17bfa41"}]},"branch":"refs/heads/master"},"0b9ae66898bc3f9487c6cf9cd51b7dbdabc97288":{"kind":"REWORK","_number":9,"created":"2023-06-16 18:31:57.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/32/881032/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/881032/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/881032/9"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-04-20 17:10:24.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-06-16 18:30:40.000000000","tz":0},"subject":"DNM: Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak","message":"DNM: Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak\n\nThis patch adds a periodic job to test restoring backups.\n\nDepends-On: Iff8e1e90ab3c7b519819577ec3aafff838e6934f\nChange-Id: Ibf2b50a23bff926a21c07ba2cc00ed16d956cf54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0b9ae66898bc3f9487c6cf9cd51b7dbdabc97288"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/0b9ae66898bc3f9487c6cf9cd51b7dbdabc97288"}]},"branch":"refs/heads/master"},"b7a42f1852356832ead3bc087621ede323ca9b18":{"kind":"REWORK","_number":10,"created":"2023-06-20 17:51:31.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/32/881032/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/881032/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/881032/10"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-04-20 17:10:24.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-06-20 17:51:09.000000000","tz":0},"subject":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak","message":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak\n\nThis patch adds a periodic job to test restoring backups.\n\nDepends-On: Iff8e1e90ab3c7b519819577ec3aafff838e6934f\nChange-Id: Ibf2b50a23bff926a21c07ba2cc00ed16d956cf54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b7a42f1852356832ead3bc087621ede323ca9b18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/b7a42f1852356832ead3bc087621ede323ca9b18"}]},"branch":"refs/heads/master"},"e62bed3ece9a356d6884bd9f0298c80ae672555c":{"kind":"REWORK","_number":11,"created":"2023-06-20 17:53:39.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/32/881032/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/881032/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/881032/11"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-04-20 17:10:24.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-06-20 17:53:23.000000000","tz":0},"subject":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak","message":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak\n\nThis patch adds a periodic job to test restoring backups.\n\nDepends-On: Iff8e1e90ab3c7b519819577ec3aafff838e6934f\nChange-Id: Ibf2b50a23bff926a21c07ba2cc00ed16d956cf54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e62bed3ece9a356d6884bd9f0298c80ae672555c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e62bed3ece9a356d6884bd9f0298c80ae672555c"}]},"branch":"refs/heads/master"},"134c62f20545206691c3b75e1de6bf53ff7c689a":{"kind":"REWORK","_number":12,"created":"2023-06-23 14:48:16.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/32/881032/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/881032/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/881032/12"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-04-20 17:10:24.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-06-23 14:48:01.000000000","tz":60},"subject":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak","message":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak\n\nThis patch adds a periodic job to test restoring backups.\n\nDepends-On: Iff8e1e90ab3c7b519819577ec3aafff838e6934f\nChange-Id: Ibf2b50a23bff926a21c07ba2cc00ed16d956cf54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/134c62f20545206691c3b75e1de6bf53ff7c689a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/134c62f20545206691c3b75e1de6bf53ff7c689a"}]},"branch":"refs/heads/master"},"4338ec07f7808c28116763f451841da714b1ff87":{"kind":"REWORK","_number":13,"created":"2023-06-27 13:35:13.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/32/881032/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/881032/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/881032/13"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-04-20 17:10:24.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-06-27 13:35:05.000000000","tz":0},"subject":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak","message":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak\n\nThis patch adds a periodic job to test restoring backups.\n\nChange-Id: Ibf2b50a23bff926a21c07ba2cc00ed16d956cf54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4338ec07f7808c28116763f451841da714b1ff87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/4338ec07f7808c28116763f451841da714b1ff87"}]},"branch":"refs/heads/master"},"77d3bf8ee47f8659eab0fd23fb23012e485832b9":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2023-06-28 10:39:21.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/32/881032/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/881032/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/881032/14"}}},"commit":{"parents":[{"commit":"d7ae9610d765919660a9f7a8769478f0b6e0aadf","subject":"Merge \"Remove six from nimble volume driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/d7ae9610d765919660a9f7a8769478f0b6e0aadf"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-04-20 17:10:24.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-06-28 10:39:21.000000000","tz":0},"subject":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak","message":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak\n\nThis patch adds a periodic job to test restoring backups.\n\nDepends-On: Iff8e1e90ab3c7b519819577ec3aafff838e6934f\nChange-Id: Ibf2b50a23bff926a21c07ba2cc00ed16d956cf54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77d3bf8ee47f8659eab0fd23fb23012e485832b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77d3bf8ee47f8659eab0fd23fb23012e485832b9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"77d6d9c6f05c9eff139634334537d9b906eb7097":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2023-06-28 10:39:27.000000000","uploader":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"ref":"refs/changes/32/881032/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/32/881032/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/32/881032/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/32/881032/15"}}},"commit":{"parents":[{"commit":"e673bcc368d3a24ec21713adbd83b4ab6cbcae18","subject":"Merge \"Remove six from solidfire driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/e673bcc368d3a24ec21713adbd83b4ab6cbcae18"}]}],"author":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-04-20 17:10:24.000000000","tz":0},"committer":{"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","date":"2023-06-28 10:39:27.000000000","tz":0},"subject":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak","message":"Zuul: Adding Periodic Job with LVM c-vol and Ceph c-bak\n\nThis patch adds a periodic job to test restoring backups.\n\nDepends-On: Iff8e1e90ab3c7b519819577ec3aafff838e6934f\nChange-Id: Ibf2b50a23bff926a21c07ba2cc00ed16d956cf54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77d6d9c6f05c9eff139634334537d9b906eb7097"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/77d6d9c6f05c9eff139634334537d9b906eb7097"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
