)]}'
{"id":"openstack%2Fcinder~881188","triplet_id":"openstack%2Fcinder~master~I6a33e4c85057bdf79530f9c9f24d867c49931310","project":"openstack/cinder","branch":"master","topic":"bug/2017815","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-09-13 14:43:04.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"35075":{"account":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"last_update":"2023-09-13 17:28:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6a33e4c85057bdf79530f9c9f24d867c49931310","subject":"Fix Infinidat driver to inherit compression","status":"MERGED","created":"2023-04-27 16:04:14.000000000","updated":"2023-11-22 18:27:35.000000000","submitted":"2023-09-13 17:28:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"881188-bug/2017815","meta_rev_id":"ce6cc4ecc72f42f0468461e93a867e5de28010d9","_number":881188,"virtual_id_number":881188,"owner":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-13 17:28:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"recommended":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"all":[{"value":1,"date":"2023-06-08 05:10:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":2,"date":"2023-09-13 14:43:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2023-09-13 08:45:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},{"value":1,"date":"2023-07-05 04:21:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":1,"date":"2023-09-13 13:45:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-12 09:54:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":1,"date":"2023-09-13 14:43:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},{"value":0,"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-27 16:30:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-27 16:36:29.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"CC"},{"updated":"2023-04-27 16:52:13.000000000","updated_by":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"reviewer":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-27 17:32:38.000000000","updated_by":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"reviewer":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"state":"CC"},{"updated":"2023-04-27 17:33:01.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-27 18:19:17.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-04-27 18:26:35.000000000","updated_by":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"reviewer":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"state":"CC"},{"updated":"2023-04-27 19:13:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-27 19:40:57.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-27 19:44:18.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-27 23:49:50.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2023-05-01 12:59:53.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-01 14:40:28.000000000","updated_by":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"reviewer":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"state":"CC"},{"updated":"2023-05-01 15:58:53.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-05-01 16:04:51.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-05-01 16:14:02.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-01 16:34:05.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-05-01 18:24:06.000000000","updated_by":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"reviewer":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"state":"CC"},{"updated":"2023-05-01 18:44:05.000000000","updated_by":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"reviewer":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"state":"CC"},{"updated":"2023-05-02 01:08:47.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-03 17:27:03.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-06-08 05:10:11.000000000","updated_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"reviewer":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"state":"REVIEWER"},{"updated":"2023-07-05 04:21:04.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"},{"updated":"2023-07-06 11:00:08.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2023-09-12 09:54:17.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2023-09-13 08:45:41.000000000","updated_by":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"reviewer":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"state":"REVIEWER"},{"updated":"2023-09-13 13:45:28.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-11-22 18:27:35.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"e1110b264020badccbc89b76175cc2f0207f8544","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-04-27 16:04:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ed6406769b24cfe8cb84fd372fadc2f9f8d1fd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 16:30:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/875ca22da3614414b0df962c7d69bc79\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ee7ae99d6ed04ff19382c585ed904c83 : SUCCESS in 24m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9bac865ad8bc45d7ad1972213b236e24 : SUCCESS in 11m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9498df7991eb807bff09812a3e75ac735c21cbea","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-04-27 16:36:29.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://127.0.0.1:9000/t/openstack/buildset/e72c2d3e8dd348deb5c7ce95e970e7ba\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/0e1b66006ccd444daf62a9c9dc46f117 : FAILURE in 27m 36s","accounts_in_message":[],"_revision_number":1},{"id":"20a9026ea156b67239e7dd99d3d80f19648b2f66","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-04-27 16:52:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/88/881188/1/ : POST_FAILURE in 39m 57s","accounts_in_message":[],"_revision_number":1},{"id":"124ced0e45d18dcbd1cb67e2bc808c6b6273e2f8","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-04-27 17:32:38.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/88/881188/1/check/iscsi-driver-alletra-6k/c7a9d88/c7a9d88.html : FAILURE in 1h 27m 09s","accounts_in_message":[],"_revision_number":1},{"id":"4e630ce13f1ef6ba4e58bba3dd2c9d8bc4b502e2","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-04-27 17:33:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/881188 : SUCCESS in 1h 26m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8e0905396e07b3d51e2ad2afd766a91617f33865","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-04-27 18:19:17.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-88-881188-1/FC/ : SUCCESS in 2h 14m 41s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-88-881188-1/iSCSI/ : SUCCESS in 1h 33m 31s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"52bc0d30161f04796045611118a5a0964c3c3471","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-04-27 18:26:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/22811ab0c33f45919bdd0470fe492e95 : POST_FAILURE in 1h 52m 16s","accounts_in_message":[],"_revision_number":1},{"id":"395be47af390a316e3a58ffd33e971514a06e8a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 19:13:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/439967e423d14603852bf07502af9b58\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3556af6fd7b4d478b92ecd24878c1e8 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9aa2d34b783d4e16ac2bf8323a9aa4f2 : SUCCESS in 13m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10c9d386a3fd467ba45aeadd8b016bed : SUCCESS in 10m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c3843fabd4a44eea5942b695c1dafc6 : SUCCESS in 12m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/c8f7b1e2c8274287aa101708362e755e : SUCCESS in 1h 02m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1060ff57e4ed4805a17486509ef60d86 : SUCCESS in 1h 39m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0d80a6ec3fb442e8863932fbd4d1789 : SUCCESS in 1h 17m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42c8a54c90084148897fd90439eff992 : SUCCESS in 11m 17s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/4a48a805e9b9425e991e840b938a9f8f : SUCCESS in 25m 39s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/f33ba7485aec498aaa6e2970a2a08783 : SUCCESS in 5m 48s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ccc44d0e89ff4c2a95d0e4ef39ef231e : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e976b59f6ab9485aa5fc395536badc31 : SUCCESS in 7m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cfb49d72380f4feea7cc8e4e3a249264 : SUCCESS in 6m 29s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5b71a18785cd4f128f6a020b4271b304 : SUCCESS in 54m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/37aad844315e4e38b015418249b5e631 : SUCCESS in 5m 18s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e7173984f110444ab1cb3561d4d56fba : FAILURE in 1h 45m 34s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/765941411b644f4fb212f1f7e9164346 : FAILURE in 1h 25m 55s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a3b57b6c6dba4625a9661ffce1685243 : SUCCESS in 2h 14m 46s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/18e233915a9043aaa5ab8c0caf2d867d : SUCCESS in 2h 03m 07s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/d60d8da826a34a72b6bf8ab64111699e : SUCCESS in 1h 08m 35s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/80a2943a5aeb4e189902dca7fbf6dcb2 : SUCCESS in 1h 27m 36s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/64af70da2b984beaa012620c6c4b83ec : SUCCESS in 42m 08s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/c9f1efd2930b4413b77996bf8b20350c : SUCCESS in 1h 20m 45s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/a72653c454ee46738f4b15df762d1888 : SUCCESS in 1h 32m 14s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/390feaaf5e8e43788d49ec9fc88e687c : SUCCESS in 2h 35m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d53548d4670c439798aed2fcee5771cb : SUCCESS in 53m 57s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/526e4923aefa4901838007e50488e17b : SUCCESS in 52m 45s","accounts_in_message":[],"_revision_number":1},{"id":"afb389e22c0393f997d44a01365b084afb905994","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-04-27 19:40:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/88/881188/1/20230428011043/NEC-FC : SUCCESS in 3h 14m 53s\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/88/881188/1/20230428011043/NEC-ISCSI : SUCCESS in 3h 30m 11s","accounts_in_message":[],"_revision_number":1},{"id":"5887df404276fb75c792392349d314eb035f38a2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-04-27 19:44:18.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-88-881188-1/ns5_nfs : SUCCESS in 3h 35m 53s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-88-881188-1/ns5_iscsi : SUCCESS in 3h 39m 36s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"814b2da1d1c2750daa9f09b18f6f0614eeb32ce1","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-04-27 23:49:50.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/88/881188/1/20230428023108/NEC-V-ISCSI : FAILURE in 5h 46m 09s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/88/881188/1/20230428023108/NEC-V-FC : FAILURE in 6h 18m 39s","accounts_in_message":[],"_revision_number":1},{"id":"7e6aa15edc99687e19dc35a635e7b0941f180e34","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-05-01 12:09:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da417f8f09727190b6729ce67b8d3e69360c2524","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-01 12:51:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/876a0767937b4ec88ae3fc97499200d1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0d31f7c385542b7ae8336a539733f16 : SUCCESS in 13m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e9e9b024096e4175989d5122786601db : FAILURE in 37m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"151d94efeaa047bab32236e30597c118e68389e3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-05-01 12:59:53.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/c98a748bb45744d0983be3749188371e : RETRY_LIMIT in 14s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/975c9d1d01ac4cbdab94b512673a9941 : RETRY_LIMIT in 1m 14s","accounts_in_message":[],"_revision_number":1},{"id":"3e15b164f8330c5a292cfc38c62f6728bc9349e8","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-01 13:03:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/e6b2dfd050674fdfbca659e487853b87 : POST_FAILURE in 52m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f4928ceeb680d38abb96849dc1d7118d1fa26c33","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-01 13:21:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/069d7d2fa53247b4ac39424d494ca3d2\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/eb57dcf894454611ad8c95ca92ca5e8b : SUCCESS in 1h 03m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0de585ddcd6633cb4232ba6d5d109d57768dc168","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-01 14:38:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1031e22dc954967831438b2bbcdf4ae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/319fd9da62a341f885c02cfa508f4ba7 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a54573069c504329acdbc98f27be5e8b : SUCCESS in 11m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16eb04ca31654a1bb3ddef256a948b24 : SUCCESS in 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be0372d3e60b45ec800483c58c4c1f20 : SUCCESS in 10m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/a2a9ee9b4cfd486c84d1bbd773452e20 : SUCCESS in 43m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5971ca9ad2a94256bd9efcc2d9d6b883 : SUCCESS in 1h 14m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a6fd8e9bf90b4c3f8321c9d318a3626b : SUCCESS in 42m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe2cc5e14684434aad7b13eca951fe99 : SUCCESS in 6m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/d9badf1180e1469faed7b0567be82da5 : SUCCESS in 16m 31s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/40594c630f11454c9806ef594bc38d11 : SUCCESS in 4m 20s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2d4c043bb9df49f1b66eec265682326b : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2b68e57d3a9d4043b0f7eb7d7779268a : SUCCESS in 5m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8232c673061b42228a320524936436c7 : SUCCESS in 5m 35s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/0ca49a459855469991bda78dbd6876da : SUCCESS in 36m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e7ca287a9ce04471b187dfc8be14ee9b : SUCCESS in 5m 52s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c3461e8f8cbc4d81b5c28f64db8ecff0 : SUCCESS in 1h 02m 17s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/4fdefaaaa1774aad8b193d3a2a18c939 : FAILURE in 1h 12m 42s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/07572b682ce04b16a834c1b6647ede7f : SUCCESS in 1h 57m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/cce73836171345c08c73b217768573d9 : RETRY_LIMIT in 3m 57s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/996d894a5b1a4c4a8525abb6fe5d1177 : SUCCESS in 48m 01s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/fed9dc0edd944101a85fcd1205ef0100 : SUCCESS in 1h 06m 38s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/91af1ae225924940abe5341689451a35 : SUCCESS in 46m 53s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/96e52e1031e440ffa1e1f44d748ae73a : SUCCESS in 56m 29s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/6892ef3680694beba21beb83f6485cb3 : RETRY_LIMIT in 2m 51s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/55445e07939d42ffa074bc4cccc4c2ca : SUCCESS in 2h 23m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e255a5ca77e4f5fbf7cd379ffe0c54b : SUCCESS in 36m 41s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17469d6d1f444352a014ba4a8f372b7a : SUCCESS in 1h 16m 35s","accounts_in_message":[],"_revision_number":1},{"id":"90ee4bf01dc679bc2a0f52466af3265d1faa014e","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-05-01 14:40:28.000000000","message":"Patch Set 1:\n\nBuild failed.","accounts_in_message":[],"_revision_number":1},{"id":"17b2cc817b46b5b8f433cbf7e0db6f47507ca442","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-01 15:58:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/88/881188/1/upstream-check/cinder-tempest-plugin-ontap-fcp/7658f38/ : SUCCESS in 1h 03m 11s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/88/881188/1/upstream-check/cinder-tempest-plugin-ontap-iscsi/1c0f020/ : SUCCESS in 45m 50s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/88/881188/1/upstream-check/cinder-tempest-plugin-ontap-nfs/7c6acf4/ : SUCCESS in 1h 19m 37s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/88/881188/1/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/4c605a3/ : FAILURE in 1h 16m 35s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"1d7eb6d90468661fee40c8ef104b67740b39bf40","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-05-01 16:04:51.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/88/881188/1/check/DellEMC_PowerStore_iSCSI/a4d9b82/DellEMC_PowerStore_iSCSI/8 : SUCCESS in 1h 14m 15s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/88/881188/1/check/DellEMC_PowerStore_FC/594d7a9/DellEMC_PowerStore_FC/9 : SUCCESS in 1h 25m 01s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/88/881188/1/check/DellEMC_PowerStore_NFS/9c5b396/DellEMC_PowerStore_NFS/7 : SUCCESS in 1h 12m 34s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/88/881188/1/check/DellEMC_PowerStore_NVMeTCP/ecb503d/DellEMC_PowerStore_NVMeTCP/7 : SUCCESS in 1h 25m 56s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"f4794044f349595d61a9eac38b59040a4a17414b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-05-01 16:14:02.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/88/881188/1/check/DellEMC_PowerFlex/a9b93e5/DellEMC_PowerFlex/47 : SUCCESS in 1h 35m 16s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"59fe0314680b6e439a95378ceb998fe04b8d3091","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-01 16:34:05.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/88/881188/1/check/DellEMC_Unity_FC/2573085/DellEMC_Unity_FC/80 : SUCCESS in 1h 36m 33s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/88/881188/1/check/DellEMC_Unity_iSCSI/eb143f1/DellEMC_Unity_iSCSI/83 : SUCCESS in 1h 55m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"41bd14d918ca19caeb93c3a345ca87e7edc19e63","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-05-01 18:24:06.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- HPMSA-iSCSI https://os-logs.seagate.com/hpmsa-ci/881188-1/HPMSA-iSCSI/ : FAILURE in 1:40:02\n- HPMSA-FC https://os-logs.seagate.com/hpmsa-ci/881188-1/HPMSA-FC/ : SUCCESS in 1:15:30","accounts_in_message":[],"_revision_number":1},{"id":"6295fbfc158ef05cabb4c7a4710c5a58c75d4a49","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-05-01 18:44:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/881188-1/Seagate-iSCSI/ : SUCCESS in 1:49:52\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/881188-1/Seagate-FC/ : SUCCESS in 1:26:06","accounts_in_message":[],"_revision_number":1},{"id":"f744437bb1008563155021e26dc7919862445390","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-05-02 01:08:47.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/88/881188/1/check/DellEMC_PowerMAX_FC/aedb69e/DellEMC_PowerMAX_FC/37 : SUCCESS in 5h 16m 17s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/88/881188/1/check/DellEMC_PowerMAX_iSCSI/89193b2/DellEMC_PowerMAX_iSCSI/37 : ABORTED in 10h 30m 01s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"aa9eb2bdf587e419d89a82c16d3755beb48874cd","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-05-03 17:27:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c912e50da3f554d36eea27d751af8675ffa74cb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-05-04 09:09:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"86ac9d8f963c2be4df142b2c44f26284db938ca1","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-05-04 09:14:17.000000000","message":"Patch Set 2:\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c5ba554273a3da5ca51f26e324707bd7c8de242b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 09:25:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/390937502887452ba5153c1bb42bf774\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b13a494f0125419c8587601545a40989 : SUCCESS in 14m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8a0b3b3deec24cc4bd0ca2a90c87a9fe : SUCCESS in 12m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b5f1851f192b0ab62d1070f85a8de36d42f35154","author":{"_account_id":31868,"name":"Open-E CI","email":"open-e-ci-joviandss@open-e.com","username":"open-e-ci-joviandss","tags":["SERVICE_USER"]},"date":"2023-05-04 09:59:38.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-joviandss-driver-ci http://51.89.139.192/openstack/cinder/88/881188/2/ : POST_FAILURE in 41m 47s","accounts_in_message":[],"_revision_number":2},{"id":"a1e956aa038de5838fbcc12e2de542015faa0a19","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-04 10:02:56.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/b62f780a9c824f50a160b37c1c700174 : POST_FAILURE in 52m 39s","accounts_in_message":[],"_revision_number":2},{"id":"1ced337162b1dd46491a93757dd6b73ec7f1a878","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-04 10:18:00.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/53dea8ee8b2d40cf80496381ee119c15\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/3149b6b2860f4d9597ad273ea3e97074 : SUCCESS in 1h 05m 26s","accounts_in_message":[],"_revision_number":2},{"id":"3ba5a5d871339dce5eaf1f1be120a766a794492b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2023-05-04 10:38:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- 3par-iscsi-driver-master-client-pip-ssa02-dsvm https://github.com/hpe-storage/hpe_cinder_logs/tree/master/new_logs/881188 : FAILURE in 1h 26m 56s","accounts_in_message":[],"_revision_number":2},{"id":"3ff3f022e32c94f1bfffcc66ec450d4cc1240526","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-04 11:09:17.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/88/881188/2/check/DellEMC_Unity_FC/bf05a33/DellEMC_Unity_FC/92 : SUCCESS in 1h 34m 25s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/88/881188/2/check/DellEMC_Unity_iSCSI/3fb535b/DellEMC_Unity_iSCSI/95 : SUCCESS in 1h 59m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"dd06f7403c9d72ec2c617a04e731c5e2319bbd47","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-05-04 11:26:48.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat FC http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-88-881188-2/FC/ : SUCCESS in 2h 5m 53s\n* Infinidat iSCSI http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-88-881188-2/iSCSI/ : SUCCESS in 2h 17m 0s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"f626e10054581a2a7e7fe9dea0a29447804a4794","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-04 11:27:59.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-alletra-6k https://hpe-storage.github.io/alletra-6k-ci-logs/logs/88/881188/2/check/iscsi-driver-alletra-6k/4a7ad0d/4a7ad0d.html : FAILURE in 2h 17m 18s","accounts_in_message":[],"_revision_number":2},{"id":"7806b961997f4facf50af04e42c9ff5bc9bd90d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 11:48:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/465bb0097aaa476abc61e28891317fdf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83ac44f756ca465eb75c897dd2075234 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5963e2e078e49a98fb4b5464b6fcfd7 : SUCCESS in 12m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb3120588d7848b69ee1ecae5c7ce649 : SUCCESS in 11m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ff76ba4b224430cb1dc0e098aef1160 : SUCCESS in 12m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/3e8042cc125149a69f1b183c4e27a73a : FAILURE in 48m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9d527d1f06f645bb820186e4d1d20168 : SUCCESS in 1h 35m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11a170492ab74131967f75ebdb840815 : SUCCESS in 1h 04m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64f69a57ec1046a89208823f9cc59fd8 : SUCCESS in 8m 57s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9378e76f326546bfb5ac16e5d9c8fc47 : SUCCESS in 17m 10s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/684d6fd14070486e9b460db2e55407c3 : SUCCESS in 5m 38s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8d51ac7c05644b9792a6442841ff088f : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/69e2b453dd0f4ae0aa168ed33cdcd4fb : SUCCESS in 6m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/edd12222c6004fc8b4dfb165f554302e : SUCCESS in 6m 22s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/5d34cd56a65643ac97d45a570b7d78c4 : SUCCESS in 36m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b5637f0ef3aa41bdbc81ca412d75cb8d : SUCCESS in 5m 51s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/302e61dfcf5849c5ad6b996269b145ba : SUCCESS in 1h 25m 40s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/61f9d9d45bca4f4fb6116116b6b983aa : FAILURE in 1h 27m 10s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4eb2085d1a81468185e7982e41c5b2d2 : SUCCESS in 1h 25m 35s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/566fef8ea0f04ad6a5120ba679f5373c : SUCCESS in 2h 27m 34s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/983109306f38413087380833ae70f8f9 : SUCCESS in 1h 50m 37s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/b741f00851b7472cb42ca88cc72388f0 : FAILURE in 1h 36m 27s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/a495d3c689b14d46ae39238af45848ab : SUCCESS in 30m 37s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/69e02a71f8eb4c22b4e877d14c023465 : SUCCESS in 1h 12m 16s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/9d396616e4064a6aa0e3f21f8450891a : SUCCESS in 1h 33m 24s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ba6e27e02c71457d82392bd40a5dd6d0 : SUCCESS in 2h 29m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e8f66118a54482386afe2fdc7157d53 : SUCCESS in 1h 00m 05s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8651505dc893426c99167c50f24f96df : SUCCESS in 1h 11m 27s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/7cd2fa76deef49119494233f9a186d26 : SUCCESS in 12m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dbd05a3b33b77a8bedf9b438e95487e96c101e3f","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-05-04 11:55:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f4b1b02d853510518303c4870f95babe828a05d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 12:07:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fb195b9f23b44fcbe3a42e2f95c7dd4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca04eb5633d74808860201f9876e9e75 : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a4559aa05d284fa49f915664a3a0a352 : SUCCESS in 9m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"52d728d622a7cd200ec11a50a4bfaf10986d0d68","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-05-04 12:47:58.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Cinder_NFS http://104.254.242.106/refs-changes-88-881188-2/ns5_nfs : SUCCESS in 3h 36m 0s\n* NexentaStor5_Cinder_iSCSI http://104.254.242.106/refs-changes-88-881188-2/ns5_iscsi : SUCCESS in 3h 37m 58s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"1996cacd401dcd5779b41486ccc910f6e5d08740","author":{"_account_id":35100,"name":"Fungible Storage CI","email":"aneeshp@fungible.com","username":"aneeshpfungible"},"date":"2023-05-04 12:58:52.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- run-devstack-cinder https://fungiblestorageci.westus.cloudapp.azure.com/zuul/t/local/build/ae7d63ae9ae147c8b5112177c302dab8 : POST_FAILURE in 1h 02m 16s","accounts_in_message":[],"_revision_number":2},{"id":"fd62977a3c555bd8133754184af5f37b539e16e5","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2023-05-04 13:02:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://127.0.0.1:9000/t/openstack/buildset/f0f5d6bd4cc44bd58e8b49af39f4c14d\n\n- dsvm-tempest-cisco-zonemanager-job https://127.0.0.1:9000/t/openstack/build/753d86b2301c4467a475a406d40772f8 : SUCCESS in 1h 03m 51s","accounts_in_message":[],"_revision_number":2},{"id":"e2121bc5b9b691008c83c79eba601b3c408baf63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 14:11:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a53817c74ef46b1a922614ce274ec2a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cdde39ad7e441608c17030c93685a4b : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4d617094fc446a1bfa62eb50f412c94 : SUCCESS in 11m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c009fdb327cd492c85bc2c08d2381886 : SUCCESS in 10m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16c3e9dc386f41b8b08a99a1a13c825a : SUCCESS in 11m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/1f505cf921a444d3ad122cdd898b1762 : SUCCESS in 54m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4b1488eaadf349558c8c381912532688 : SUCCESS in 1h 29m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/874a51882b664bfebee3a8e445e24960 : SUCCESS in 57m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e27e22737aa4f529e9e169c8f1e1e4d : SUCCESS in 10m 34s\n- cinder-code-coverage https://zuul.opendev.org/t/openstack/build/9523ef1659da45eaafc830ac1fe67076 : SUCCESS in 16m 36s (non-voting)\n- cinder-mypy https://zuul.opendev.org/t/openstack/build/debe622b1832490aa32489f01cad2d6b : SUCCESS in 5m 56s\n- cinder-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/93b5716611114b20bbfecadee445a642 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/59e3a3dbed41468dabd4480347374139 : SUCCESS in 6m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c21d5778eb6e498f85ccd9349e2f9291 : SUCCESS in 6m 09s\n- cinder-rally-task https://zuul.opendev.org/t/openstack/build/6f27efea031c489b867763f69e14167b : SUCCESS in 57m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/854c825d094e4cdba07ce8c6a412b14b : SUCCESS in 6m 03s (non-voting)\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/374ca27c0c8b439e8ee986c4f3685463 : SUCCESS in 1h 15m 24s\n- cinder-plugin-ceph-tempest-mn-aa https://zuul.opendev.org/t/openstack/build/dae3c68e0cf148a8ab459842f106277f : FAILURE in 1h 32m 14s (non-voting)\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8a0ca46b38404cd29a38a9f41d1c6626 : SUCCESS in 1h 39m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-fips https://zuul.opendev.org/t/openstack/build/f06b24d7de5f4ad5a3c04a6c99ee4b18 : SUCCESS in 2h 10m 23s (non-voting)\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/984b1ac1483c4b728a73834afd963f02 : SUCCESS in 1h 08m 18s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/0ee15085bcb34e98b61486a9a4c4e5c4 : SUCCESS in 1h 46m 32s (non-voting)\n- cinder-for-glance-optimized https://zuul.opendev.org/t/openstack/build/c79b29c2c26a4e15a06b199b20ce0b9a : SUCCESS in 45m 21s (non-voting)\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/1a39246deb064b71a5e84d4c987bf181 : SUCCESS in 1h 14m 58s\n- devstack-plugin-nfs-tempest-full-fips https://zuul.opendev.org/t/openstack/build/74bd585885d843868dcc7a1dfd5aec73 : SUCCESS in 1h 42m 55s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1bf54842b8754f789f853feb2e381403 : SUCCESS in 2h 10m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2a67cf43a78449e86db027fb46bef68 : SUCCESS in 53m 18s\n- tempest-integrated-storage-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6a4d8758e3e45939388d5f3017cc3ca : SUCCESS in 1h 12m 18s\n- cinder-tox-py311 https://zuul.opendev.org/t/openstack/build/7311b04ed21a467aacdaa6211d7d6dbc : SUCCESS in 10m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"681e2bb51a7e72b65a816b381348e220203d123d","tag":"autogenerated:hpmsa-ci","author":{"_account_id":24814,"name":"HPMSA CI","email":"cinder+hpmsa@tristero.net","username":"hpmsa-ci"},"date":"2023-05-04 14:13:00.000000000","message":"Patch Set 2:\n\nBuild failed.","accounts_in_message":[],"_revision_number":2},{"id":"f85eda5b8a5ea2924f0f9e7aa197b4be062a5686","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2023-05-04 15:17:20.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/88/881188/2/20230504180933/NEC-V-ISCSI : SUCCESS in 2h 28m 47s\n- NEC-V-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/88/881188/2/20230504180933/NEC-V-FC : SUCCESS in 6h 07m 43s","accounts_in_message":[],"_revision_number":2},{"id":"4adc8d041a95fea1d62add9965f8476ae573773c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-04 15:30:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-tempest-plugin-ontap-fcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/88/881188/2/upstream-check/cinder-tempest-plugin-ontap-fcp/7974788/ : SUCCESS in 1h 03m 57s\n- cinder-tempest-plugin-ontap-iscsi https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/88/881188/2/upstream-check/cinder-tempest-plugin-ontap-iscsi/69056a7/ : SUCCESS in 1h 14m 42s\n- cinder-tempest-plugin-ontap-nfs https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/88/881188/2/upstream-check/cinder-tempest-plugin-ontap-nfs/bcb7a5e/ : SUCCESS in 1h 14m 45s\n- cinder-tempest-plugin-ontap-nvme-tcp https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/88/881188/2/upstream-check/cinder-tempest-plugin-ontap-nvme-tcp/f3b084a/ : SUCCESS in 1h 17m 49s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"21e263cc0272f89bda07ea2efffe8919fd74905c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-05-04 15:40:49.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex https://elab-os-logsrv.delllabs.net/88/881188/2/check/DellEMC_PowerFlex/9a3cb33/DellEMC_PowerFlex/56 : SUCCESS in 1h 28m 58s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"4c5c6a3c8e82a9aac8273f2f6b98e5186b6bb30c","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-05-04 15:41:09.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_iSCSI https://elab-os-logsrv.delllabs.net/88/881188/2/check/DellEMC_PowerStore_iSCSI/9329efe/DellEMC_PowerStore_iSCSI/17 : SUCCESS in 1h 12m 36s\n- DellEMC_PowerStore_FC https://elab-os-logsrv.delllabs.net/88/881188/2/check/DellEMC_PowerStore_FC/ced0412/DellEMC_PowerStore_FC/18 : SUCCESS in 1h 29m 16s\n- DellEMC_PowerStore_NFS https://elab-os-logsrv.delllabs.net/88/881188/2/check/DellEMC_PowerStore_NFS/b658805/DellEMC_PowerStore_NFS/16 : SUCCESS in 1h 17m 32s\n- DellEMC_PowerStore_NVMeTCP https://elab-os-logsrv.delllabs.net/88/881188/2/check/DellEMC_PowerStore_NVMeTCP/24e9f7f/DellEMC_PowerStore_NVMeTCP/16 : SUCCESS in 1h 26m 17s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"5ecfc8a5bfd1bbba30a89c46dff1975522005971","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-04 16:27:05.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_FC https://elab-os-logsrv.delllabs.net/88/881188/2/check/DellEMC_Unity_FC/97c85a7/DellEMC_Unity_FC/93 : SUCCESS in 1h 32m 06s\n- DellEMC_Unity_iSCSI https://elab-os-logsrv.delllabs.net/88/881188/2/check/DellEMC_Unity_iSCSI/a8c9441/DellEMC_Unity_iSCSI/96 : SUCCESS in 2h 15m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"80759133f25f9cf0678397be098af11bd1a3396c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2023-05-04 20:01:21.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://fl9-119-243-99-132.tky.mesh.ad.jp/88/881188/2/20230505012454/NEC-ISCSI : SUCCESS in 3h 08m 10s\n- NEC-FC https://fl9-119-243-99-132.tky.mesh.ad.jp/88/881188/2/20230505012454/NEC-FC : SUCCESS in 3h 36m 24s","accounts_in_message":[],"_revision_number":2},{"id":"d6e850f630e78804a8aff6abe1c7fb4a97e829a9","tag":"autogenerated:dellpvme-ci","author":{"_account_id":32619,"name":"DellEMC PowerVault ME CI","email":"cinder+dellpvme@tristero.net","username":"dellpvme-ci"},"date":"2023-05-04 20:34:20.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- PVME-iSCSI https://os-logs.seagate.com/dell-ci/881188-2/PVME-iSCSI/ : SUCCESS in 1:57:45\n- PVME-FC https://os-logs.seagate.com/dell-ci/881188-2/PVME-FC/ : SUCCESS in 1:33:24","accounts_in_message":[],"_revision_number":2},{"id":"74454fca28aa72278f5c8913afbc1b1f25fe75d5","tag":"autogenerated:seagate-ci","author":{"_account_id":30688,"name":"Seagate CI","email":"cinder+seagate@tristero.net","username":"seagate-ci"},"date":"2023-05-04 20:34:24.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n- Seagate-iSCSI https://os-logs.seagate.com/seagate-ci/881188-2/Seagate-iSCSI/ : SUCCESS in 2:02:19\n- Seagate-FC https://os-logs.seagate.com/seagate-ci/881188-2/Seagate-FC/ : SUCCESS in 1:30:48","accounts_in_message":[],"_revision_number":2},{"id":"e41c29096e447c6b68cde4b12ade6eefe1097e96","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-05-05 23:24:20.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/1122572b61ca47ea9383be8a22e9e157 : RETRY_LIMIT in 1m 52s\n- devstack-plugin-ceph-dsvm-focal-py3 finger://zuul-v3-executor.novalocal/316c69d3914c4a87aa2da0ff20e9db3c : RETRY_LIMIT in 1m 04s","accounts_in_message":[],"_revision_number":2},{"id":"f86ac7d898813ec326d2b23c552e5b8b42495124","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-05-06 08:20:33.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_FC https://elab-os-logsrv.delllabs.net/88/881188/2/check/DellEMC_PowerMAX_FC/443ec1b/DellEMC_PowerMAX_FC/44 : SUCCESS in 5h 22m 03s\n- DellEMC_PowerMAX_iSCSI https://elab-os-logsrv.delllabs.net/88/881188/2/check/DellEMC_PowerMAX_iSCSI/124db9f/DellEMC_PowerMAX_iSCSI/47 : FAILURE in 2s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"f23c436681b98648724a326f7683ae4f6bc3a5ce","author":{"_account_id":34318,"name":"HPE Nimble Storage CI","display_name":"HPE Nimble Storage CI","email":"raghavendra.tilay@capgemini.com","username":"hpe-nimble"},"date":"2023-05-10 09:28:58.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- iscsi-driver-nimble iscsi-driver-nimble : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"868ee66169c95c6b3b906ae058c5ddf2c656ae91","author":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"},"date":"2023-06-08 05:10:11.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9d24222aaacc5c7f63ab2ac8a1ca10b196e96773","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2023-07-05 04:21:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c7f6d2a0a3bfb85bfe6f436ce028576a48705ca9","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2023-07-06 11:00:08.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/88/881188/2/fc : SUCCESS in 0h 44m 33s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/88/881188/2/iscsi : SUCCESS in 1h 16m 34s","accounts_in_message":[],"_revision_number":2},{"id":"6b2cd9b0065f7211b0214cc35c20d4e7a08706e8","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-09-12 09:54:17.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6d30c57c1f688a61efc2517faf8d78900eafb41d","author":{"_account_id":36132,"name":"Cuiye Liu","display_name":"Cherry","email":"cuiye.liu@dell.com","username":"cuiyeliu"},"date":"2023-09-13 08:45:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cb2d973edd0fb6501e2cd8a7eaae6ac97fd604ed","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-09-13 13:45:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c6adeef57b85be0e9510d31ad7cf4876b4c08e11","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-09-13 14:43:04.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b5c9972c8b1bc3d49935f5f1609881537581d87b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-09-13 14:43:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"79b86ad896e75b300d10ae1074f91d61e02e249b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 14:44:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"53f9211a5aef56c05ae7ac19c703583bbea6d611","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 17:28:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d29e57cd8c8c4c9189a9af183ee9a265\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/270c5b6956304b6998aac57b5dd7cc99 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d14689df6664d75a351e26b464fc916 : SUCCESS in 11m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a2dc87d2cd347f789553d698b5be800 : SUCCESS in 12m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a667df30ec7549199aa85f1e65a89a7f : SUCCESS in 13m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3457b7cd8d64a6dad7ca88cd5f45769 : SUCCESS in 13m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/b833962072d346d9b0def8fd7e3836b8 : SUCCESS in 1h 05m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/57dce719b2dd488db3169ac2477450f4 : SUCCESS in 1h 25m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/027fe1ca764f4c1291c2e6db493b7605 : SUCCESS in 1h 00m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df78154b7f3d47cb8507b499247c7644 : SUCCESS in 8m 19s\n- cinder-grenade-mn-sub-volbak https://zuul.opendev.org/t/openstack/build/f18518f2c52e490ba6e5022f3e038fdc : SUCCESS in 58m 35s\n- cinder-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d2363c811ea74c039e6023c167202957 : SUCCESS in 1h 58m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/300158a5ab8546e58cf51c345e2c0a35 : SUCCESS in 1h 13m 29s","accounts_in_message":[],"_revision_number":2},{"id":"2bed1e466c52fbf24687d2aeeb92bc9387ce08c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 17:28:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0af68626ebc6c45e049ae777a20835926d5ba7d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 17:29:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b556641ddac4f9d8577431cb58f2ee7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1212d5f5f9843fda8ba29a893166bef : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bab51073cbd48638e4942da2cf414e5 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2},{"id":"ae877242e2ecda02549550aba08c46628e0ba5e8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-11-22 18:27:35.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":2,"current_revision":"026dd9d34a3fa9bfc912cb350953aef64a6864a5","revisions":{"6ab18cd07a008d6dbf0b3d8948e2fdf346e516a8":{"kind":"REWORK","_number":1,"created":"2023-04-27 16:04:14.000000000","uploader":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"ref":"refs/changes/88/881188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/881188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/881188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/881188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/881188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/881188/1"}}},"commit":{"parents":[{"commit":"40147e2489c1f4f9940dc3df2cfa0fa81b5070df","subject":"Merge \"Require novaclient \u003e\u003d 18.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40147e2489c1f4f9940dc3df2cfa0fa81b5070df"}]}],"author":{"name":"Alexander Deiter","email":"adeiter@infinidat.com","date":"2023-04-27 15:50:30.000000000","tz":0},"committer":{"name":"Alexander Deiter","email":"adeiter@infinidat.com","date":"2023-04-27 16:02:48.000000000","tz":0},"subject":"Fix Infinidat driver to inherit compression","message":"Fix Infinidat driver to inherit compression\n\n* Fixed Infinidat driver to inherit compression setting by default\n  for all newly created volumes. Admin can set `True` or `False`\n  for the `infinidat_use_compression` option in the driver section\n  of cinder.conf to explicitly enable or disable compression setting\n  for all newly created volumes.\n\n* Removed support for pre-v3.0 InfiniBox systems that do not support\n  compression. These versions are end of life and no longer supported\n  for a long time\n\nCloses-bug: #2017815\nSigned-off-by: Alexander Deiter \u003cadeiter@infinidat.com\u003e\nChange-Id: I6a33e4c85057bdf79530f9c9f24d867c49931310\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6ab18cd07a008d6dbf0b3d8948e2fdf346e516a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/6ab18cd07a008d6dbf0b3d8948e2fdf346e516a8"}]},"branch":"refs/heads/master"},"026dd9d34a3fa9bfc912cb350953aef64a6864a5":{"kind":"REWORK","_number":2,"created":"2023-05-04 09:09:30.000000000","uploader":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"ref":"refs/changes/88/881188/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cinder","ref":"refs/changes/88/881188/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/881188/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/881188/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cinder refs/changes/88/881188/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cinder refs/changes/88/881188/2"}}},"commit":{"parents":[{"commit":"40147e2489c1f4f9940dc3df2cfa0fa81b5070df","subject":"Merge \"Require novaclient \u003e\u003d 18.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/40147e2489c1f4f9940dc3df2cfa0fa81b5070df"}]}],"author":{"name":"Alexander Deiter","email":"adeiter@infinidat.com","date":"2023-04-27 15:50:30.000000000","tz":0},"committer":{"name":"Alexander Deiter","email":"adeiter@infinidat.com","date":"2023-05-04 08:55:12.000000000","tz":0},"subject":"Fix Infinidat driver to inherit compression","message":"Fix Infinidat driver to inherit compression\n\n* Fixed Infinidat driver to inherit compression setting by default for\n  all newly created volumes. Admin can set ``True`` or ``False`` for\n  the ``infinidat_use_compression`` option in the driver section of\n  cinder.conf to explicitly enable or disable compression setting for\n  all newly created volumes. Or leave this option unset (commented out)\n  for all created volumes to inherit their compression setting from\n  their parent pool at creation time. The default value is unset.\n\n* Removed support for pre-v3.0 InfiniBox systems that do not support\n  compression. These versions are end of life and no longer supported\n  for a long time\n\nCloses-bug: #2017815\nSigned-off-by: Alexander Deiter \u003cadeiter@infinidat.com\u003e\nChange-Id: I6a33e4c85057bdf79530f9c9f24d867c49931310\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/026dd9d34a3fa9bfc912cb350953aef64a6864a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cinder/commit/026dd9d34a3fa9bfc912cb350953aef64a6864a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29122,"name":"Raghavendra Tilay","email":"raghavendra-uddhav.tilay@hpe.com","username":"raghavendrat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
